)]}'
{"id":"openstack%2Fopenstack-ansible-lxc_hosts~894557","triplet_id":"openstack%2Fopenstack-ansible-lxc_hosts~master~Ia78429eaf4bd71a8f3509c4e484f7dd02574c6b1","project":"openstack/openstack-ansible-lxc_hosts","branch":"master","topic":"osa-bookworm","attention_set":{},"removed_from_attention_set":{"25023":{"account":{"_account_id":25023,"name":"Jonathan Rosser","email":"jonathan.rosser@rd.bbc.co.uk","username":"jrosser"},"last_update":"2023-10-10 11:16:38.000000000","reason":"Change was submitted"}},"hashtags":[],"change_id":"Ia78429eaf4bd71a8f3509c4e484f7dd02574c6b1","subject":"Add ca-certificates into debian base image during debootstrap.","status":"MERGED","created":"2023-09-11 16:14:19.000000000","updated":"2023-10-10 11:17:44.000000000","submitted":"2023-10-10 11:16:38.000000000","submitter":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"total_comment_count":3,"unresolved_comment_count":0,"has_review_started":true,"submission_id":"894557-osa-bookworm","meta_rev_id":"b925a34528c7a47edfd3e9d4c0d55f7a44f7ffee","_number":894557,"virtual_id_number":894557,"owner":{"_account_id":25023,"name":"Jonathan Rosser","email":"jonathan.rosser@rd.bbc.co.uk","username":"jrosser"},"actions":{},"labels":{"Verified":{"approved":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"all":[{"value":0,"_account_id":31542,"name":"Andrew Bonney","email":"andrew.bonney@bbc.co.uk","username":"andrewbonney"},{"value":0,"_account_id":28619,"name":"Dmitriy Rabotyagov","email":"noonedeadpunk@gmail.com","username":"noonedeadpunk"},{"tag":"autogenerated:zuul:gate","value":2,"date":"2023-10-10 11:16:38.000000000","permitted_voting_range":{"min":2,"max":2},"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]}],"values":{"-2":"Fails","-1":"Doesn\u0027t seem to work"," 0":"No score","+1":"Works for me","+2":"Verified"},"description":"","default_value":0,"optional":true},"Code-Review":{"approved":{"_account_id":31542,"name":"Andrew Bonney","email":"andrew.bonney@bbc.co.uk","username":"andrewbonney"},"all":[{"value":2,"date":"2023-10-10 08:43:21.000000000","permitted_voting_range":{"min":2,"max":2},"_account_id":31542,"name":"Andrew Bonney","email":"andrew.bonney@bbc.co.uk","username":"andrewbonney"},{"value":2,"date":"2023-10-02 06:50:46.000000000","permitted_voting_range":{"min":2,"max":2},"_account_id":28619,"name":"Dmitriy Rabotyagov","email":"noonedeadpunk@gmail.com","username":"noonedeadpunk"},{"value":0,"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]}],"values":{"-2":"Do not merge","-1":"This patch needs further work before it can be merged"," 0":"No score","+1":"Looks good to me, but someone else must approve","+2":"Looks good to me (core reviewer)"},"description":"","default_value":0,"optional":true},"Workflow":{"approved":{"_account_id":31542,"name":"Andrew Bonney","email":"andrew.bonney@bbc.co.uk","username":"andrewbonney"},"all":[{"value":1,"date":"2023-10-10 08:43:21.000000000","permitted_voting_range":{"min":1,"max":1},"_account_id":31542,"name":"Andrew Bonney","email":"andrew.bonney@bbc.co.uk","username":"andrewbonney"},{"value":0,"_account_id":28619,"name":"Dmitriy Rabotyagov","email":"noonedeadpunk@gmail.com","username":"noonedeadpunk"},{"value":0,"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]}],"values":{"-1":"Work in progress"," 0":"Ready for reviews","+1":"Approved"},"description":"","default_value":0,"optional":true},"Backport-Candidate":{"all":[{"value":0,"_account_id":31542,"name":"Andrew Bonney","email":"andrew.bonney@bbc.co.uk","username":"andrewbonney"},{"value":0,"_account_id":28619,"name":"Dmitriy Rabotyagov","email":"noonedeadpunk@gmail.com","username":"noonedeadpunk"},{"value":0,"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]}],"values":{"-1":"Not A Backport Candidate"," 0":"Backport Review Needed","+1":"Proposed Backport"},"description":"","default_value":0,"optional":true}},"removable_reviewers":[],"reviewers":{"REVIEWER":[{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},{"_account_id":28619,"name":"Dmitriy Rabotyagov","email":"noonedeadpunk@gmail.com","username":"noonedeadpunk"},{"_account_id":31542,"name":"Andrew Bonney","email":"andrew.bonney@bbc.co.uk","username":"andrewbonney"}]},"pending_reviewers":{},"reviewer_updates":[{"updated":"2023-09-11 18:24:22.000000000","updated_by":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"reviewer":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"state":"REVIEWER"},{"updated":"2023-09-29 08:10:10.000000000","updated_by":{"_account_id":28619,"name":"Dmitriy Rabotyagov","email":"noonedeadpunk@gmail.com","username":"noonedeadpunk"},"reviewer":{"_account_id":28619,"name":"Dmitriy Rabotyagov","email":"noonedeadpunk@gmail.com","username":"noonedeadpunk"},"state":"REVIEWER"},{"updated":"2023-10-10 08:43:21.000000000","updated_by":{"_account_id":31542,"name":"Andrew Bonney","email":"andrew.bonney@bbc.co.uk","username":"andrewbonney"},"reviewer":{"_account_id":31542,"name":"Andrew Bonney","email":"andrew.bonney@bbc.co.uk","username":"andrewbonney"},"state":"REVIEWER"}],"messages":[{"id":"08428727a6a68cd62322deea19b9549f7a049beb","tag":"autogenerated:gerrit:newPatchSet","author":{"_account_id":25023,"name":"Jonathan Rosser","email":"jonathan.rosser@rd.bbc.co.uk","username":"jrosser"},"date":"2023-09-11 16:14:19.000000000","message":"Uploaded patch set 1.","accounts_in_message":[],"_revision_number":1},{"id":"d96a8f8d108ba364ff2eb100bb7b5db167d9f81b","tag":"autogenerated:zuul:check","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2023-09-11 18:24:22.000000000","message":"Patch Set 1: Verified-1\n\nBuild failed (check pipeline).  For information on how to proceed, see\nhttps://docs.opendev.org/opendev/infra-manual/latest/developers.html#automated-testing\n\nhttps://zuul.opendev.org/t/openstack/buildset/87dc51f2bb3e467f8bba81b2bfcf976a\n\n- openstack-ansible-linters-ubuntu-jammy https://zuul.opendev.org/t/openstack/build/c4ac9acb495e44c99007cba8d1e1cb57 : SUCCESS in 17m 45s\n- openstack-tox-docs https://zuul.opendev.org/t/openstack/build/7ea896ce46f1439abebfbd019033b78f : SUCCESS in 5m 31s\n- openstack-ansible-deploy-infra_lxc-centos-9-stream https://zuul.opendev.org/t/openstack/build/892d6272fdb94fe8881548ce081d75a1 : SUCCESS in 1h 32m 30s (non-voting)\n- openstack-ansible-deploy-infra_lxc-rockylinux-9 https://zuul.opendev.org/t/openstack/build/ee453c9173b8459f90855ba7b7be3fa2 : SUCCESS in 1h 33m 33s\n- openstack-ansible-deploy-infra_lxc_tls-rockylinux-9 https://zuul.opendev.org/t/openstack/build/c2e7dfb8a32a4750b688d8e50dcd77b1 : SUCCESS in 1h 27m 44s\n- openstack-ansible-deploy-infra_lxc-debian-bullseye https://zuul.opendev.org/t/openstack/build/2986fcb0a8594a4599998bfa0506f64f : FAILURE in 19m 08s\n- openstack-ansible-deploy-infra_lxc-ubuntu-jammy https://zuul.opendev.org/t/openstack/build/dcf8c6993ddc470da500cdb859de5fc2 : FAILURE in 20m 56s\n- openstack-ansible-deploy-infra_lxc_stepca-ubuntu-jammy https://zuul.opendev.org/t/openstack/build/8d116d62650c440a9a1c3c24f4d2849c : FAILURE in 17m 30s\n- openstack-ansible-deploy-infra_lxc_tls-ubuntu-jammy https://zuul.opendev.org/t/openstack/build/ff305bdfdbe241cfb595e00da4fa2845 : FAILURE in 18m 54s\n- openstack-ansible-upgrade-infra_lxc-rockylinux-9 https://zuul.opendev.org/t/openstack/build/3724e010c49d4aac9caaf841990c7fc4 : SUCCESS in 2h 03m 11s\n- openstack-ansible-upgrade-infra_lxc-ubuntu-jammy https://zuul.opendev.org/t/openstack/build/bb9f6d183d3b4e7f893cda69c716aea2 : SUCCESS in 1h 54m 25s\n- openstack-ansible-deploy-hosts_lxc_zfs-ubuntu-jammy https://zuul.opendev.org/t/openstack/build/2f72cc8709c34e0d98b4e7a9e21fb031 : FAILURE in 24m 47s\n- openstack-ansible-deploy-hosts_lxc_btrfs-ubuntu-jammy https://zuul.opendev.org/t/openstack/build/516cb0f670d0466383881bc648599738 : FAILURE in 22m 46s\n- openstack-ansible-deploy-hosts_lxc_overlayfs-ubuntu-jammy https://zuul.opendev.org/t/openstack/build/7a8c6c226cef4cb380d6315444c48f0e : FAILURE in 20m 51s\n- openstack-ansible-deploy-hosts_lxc_btrfs-debian-bullseye https://zuul.opendev.org/t/openstack/build/533877fb0bf8425c8c439bb437416090 : FAILURE in 17m 34s\n- openstack-ansible-deploy-hosts_lxc_overlayfs-debian-bullseye https://zuul.opendev.org/t/openstack/build/d2fdec4520be4d4390380d2ca7cdb349 : FAILURE in 19m 42s\n- openstack-ansible-deploy-hosts_lxc_overlayfs-rockylinux-9 https://zuul.opendev.org/t/openstack/build/bd3a9f3571dd4bac8881a4026a4880a4 : SUCCESS in 34m 00s","accounts_in_message":[],"_revision_number":1},{"id":"344f3e20d6832759f58a37df2b6f486e8cc95b76","tag":"autogenerated:gerrit:newPatchSet","author":{"_account_id":25023,"name":"Jonathan Rosser","email":"jonathan.rosser@rd.bbc.co.uk","username":"jrosser"},"date":"2023-09-11 18:30:02.000000000","message":"Uploaded patch set 2: Patch Set 1 was rebased.\n\nOutdated Votes:\n* Verified-1\n","accounts_in_message":[],"_revision_number":2},{"id":"18e4b2c546d0b0a44b363a7c34788f9f3ffb5336","tag":"autogenerated:zuul:check","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2023-09-11 20:53:02.000000000","message":"Patch Set 2: Verified+1\n\nBuild succeeded (check pipeline).\nhttps://zuul.opendev.org/t/openstack/buildset/0854ce1d675d4f56bd98e16603212989\n\n- openstack-ansible-linters-ubuntu-jammy https://zuul.opendev.org/t/openstack/build/c702b45e01804764af633880c0a8158a : SUCCESS in 11m 25s\n- openstack-tox-docs https://zuul.opendev.org/t/openstack/build/686fa6b177394607a38bc9fc2a31cbfc : SUCCESS in 4m 15s\n- openstack-ansible-deploy-infra_lxc-centos-9-stream https://zuul.opendev.org/t/openstack/build/b360287f475a48f99d451f96cbea4552 : SUCCESS in 1h 33m 57s (non-voting)\n- openstack-ansible-deploy-infra_lxc-rockylinux-9 https://zuul.opendev.org/t/openstack/build/9deefb06f38945cda03ea44fe000822c : SUCCESS in 1h 27m 08s\n- openstack-ansible-deploy-infra_lxc_tls-rockylinux-9 https://zuul.opendev.org/t/openstack/build/63f1e06ad7224911bf1acf0cbd11a2ba : SUCCESS in 1h 33m 07s\n- openstack-ansible-deploy-infra_lxc-debian-bullseye https://zuul.opendev.org/t/openstack/build/c3c048ebdf1042f7bcff9c9e9822a735 : SUCCESS in 1h 21m 45s\n- openstack-ansible-deploy-infra_lxc-ubuntu-jammy https://zuul.opendev.org/t/openstack/build/5e3647fa90ed4c909a42e6d060632e72 : SUCCESS in 1h 20m 44s\n- openstack-ansible-deploy-infra_lxc_stepca-ubuntu-jammy https://zuul.opendev.org/t/openstack/build/0f9440f672ec4e04931d23086a634a81 : SUCCESS in 1h 23m 32s\n- openstack-ansible-deploy-infra_lxc_tls-ubuntu-jammy https://zuul.opendev.org/t/openstack/build/8f452acbd9bb4623b772d64687a8bbdc : SUCCESS in 1h 10m 29s\n- openstack-ansible-upgrade-infra_lxc-rockylinux-9 https://zuul.opendev.org/t/openstack/build/50025b00c76442469ced0e453e3ffe2f : SUCCESS in 2h 13m 27s\n- openstack-ansible-upgrade-infra_lxc-ubuntu-jammy https://zuul.opendev.org/t/openstack/build/748fc793ed2e4fd3864d9cd0d283d21e : SUCCESS in 1h 50m 12s\n- openstack-ansible-deploy-hosts_lxc_zfs-ubuntu-jammy https://zuul.opendev.org/t/openstack/build/8a98e57578204cadb329ff389f98d373 : SUCCESS in 27m 38s\n- openstack-ansible-deploy-hosts_lxc_btrfs-ubuntu-jammy https://zuul.opendev.org/t/openstack/build/3b453deec7c641bd96bf2a0df1f403c1 : SUCCESS in 26m 35s\n- openstack-ansible-deploy-hosts_lxc_overlayfs-ubuntu-jammy https://zuul.opendev.org/t/openstack/build/05dacb0236e842d5be71d3976239a715 : SUCCESS in 27m 59s\n- openstack-ansible-deploy-hosts_lxc_btrfs-debian-bullseye https://zuul.opendev.org/t/openstack/build/8fc44348f99d4e6983f88b2ba7f52c08 : SUCCESS in 24m 06s\n- openstack-ansible-deploy-hosts_lxc_overlayfs-debian-bullseye https://zuul.opendev.org/t/openstack/build/eb8de6f06e154c57bd5fc3dd85fa81fa : SUCCESS in 34m 24s\n- openstack-ansible-deploy-hosts_lxc_overlayfs-rockylinux-9 https://zuul.opendev.org/t/openstack/build/0a90ba28301044118400a651f74fb9ab : SUCCESS in 32m 33s","accounts_in_message":[],"_revision_number":2},{"id":"bb11900779a75e58cb5d1ae8424c6fb4aad58d3a","tag":"autogenerated:gerrit:newPatchSet","author":{"_account_id":25023,"name":"Jonathan Rosser","email":"jonathan.rosser@rd.bbc.co.uk","username":"jrosser"},"date":"2023-09-27 16:29:37.000000000","message":"Patch Set 3: Patch Set 2 was rebased\n\nOutdated Votes:\n* Verified+1\n","accounts_in_message":[],"_revision_number":3},{"id":"9ee9d69b06ffd3026f25c36c00107d2fd0bb66a8","tag":"autogenerated:zuul:check","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2023-09-27 18:42:14.000000000","message":"Patch Set 3: Verified-1\n\nBuild failed (check pipeline).  For information on how to proceed, see\nhttps://docs.opendev.org/opendev/infra-manual/latest/developers.html#automated-testing\n\nhttps://zuul.opendev.org/t/openstack/buildset/b2af2c33452e4d97bc3c5cbfd8249451\n\n- openstack-ansible-linters-ubuntu-jammy https://zuul.opendev.org/t/openstack/build/28f571e17f084b73887f0296493299ec : FAILURE in 14m 03s\n- openstack-tox-docs https://zuul.opendev.org/t/openstack/build/85a3605956b84fd491f4d1e139a38732 : SUCCESS in 4m 44s\n- openstack-ansible-deploy-infra_lxc-centos-9-stream https://zuul.opendev.org/t/openstack/build/e543e1bb2d964e04aea247752a735686 : SUCCESS in 1h 34m 43s (non-voting)\n- openstack-ansible-deploy-infra_lxc-rockylinux-9 https://zuul.opendev.org/t/openstack/build/9ae2ea95580a46789223fcc71753ab98 : SUCCESS in 1h 30m 04s\n- openstack-ansible-deploy-infra_lxc_tls-rockylinux-9 https://zuul.opendev.org/t/openstack/build/5b971fbaad3e40c7ab27ea7a48a60b59 : SUCCESS in 1h 36m 51s\n- openstack-ansible-deploy-infra_lxc-debian-bullseye https://zuul.opendev.org/t/openstack/build/1a7a733c35fc43d584edba47577ad73b : SUCCESS in 1h 25m 41s\n- openstack-ansible-deploy-infra_lxc-ubuntu-jammy https://zuul.opendev.org/t/openstack/build/fe522485865549088742b2b60310a8ce : SUCCESS in 1h 07m 11s\n- openstack-ansible-deploy-infra_lxc_stepca-ubuntu-jammy https://zuul.opendev.org/t/openstack/build/5229ac2f6e8844089c5ec24a10db8bcc : SUCCESS in 1h 25m 45s\n- openstack-ansible-deploy-infra_lxc_tls-ubuntu-jammy https://zuul.opendev.org/t/openstack/build/b6f25c36ba3d4e7182d0dc3038bb3454 : SUCCESS in 1h 15m 10s\n- openstack-ansible-upgrade-infra_lxc-rockylinux-9 https://zuul.opendev.org/t/openstack/build/21d9e25722cd4a9ca68da6c9b3a7d634 : SUCCESS in 2h 05m 37s\n- openstack-ansible-upgrade-infra_lxc-ubuntu-jammy https://zuul.opendev.org/t/openstack/build/0c039dc91c60455fb8ee44e30449c662 : SUCCESS in 2h 02m 45s\n- openstack-ansible-deploy-hosts_lxc_zfs-ubuntu-jammy https://zuul.opendev.org/t/openstack/build/72608117450e4dc79b2c0d0ed4f11ac1 : SUCCESS in 31m 24s\n- openstack-ansible-deploy-hosts_lxc_btrfs-ubuntu-jammy https://zuul.opendev.org/t/openstack/build/f71482b3e4094c93bb4360a1d8bd7a84 : SUCCESS in 28m 31s\n- openstack-ansible-deploy-hosts_lxc_overlayfs-ubuntu-jammy https://zuul.opendev.org/t/openstack/build/2f3d8e4460ad40d884f4adffaa475f7c : SUCCESS in 27m 52s\n- openstack-ansible-deploy-hosts_lxc_btrfs-debian-bullseye https://zuul.opendev.org/t/openstack/build/e8c8cf61a5fb4d57bb3be556d7b1c343 : SUCCESS in 35m 53s\n- openstack-ansible-deploy-hosts_lxc_overlayfs-debian-bullseye https://zuul.opendev.org/t/openstack/build/14d9f166846741a9813ecea17abcee14 : SUCCESS in 26m 54s\n- openstack-ansible-deploy-hosts_lxc_overlayfs-rockylinux-9 https://zuul.opendev.org/t/openstack/build/fb2a296d9f13449cb4ddc0be94be9b40 : SUCCESS in 37m 01s","accounts_in_message":[],"_revision_number":3},{"id":"e9b698af0e8d40b088cc27a42d4722ef06c059ad","tag":"autogenerated:gerrit:newPatchSet","author":{"_account_id":25023,"name":"Jonathan Rosser","email":"jonathan.rosser@rd.bbc.co.uk","username":"jrosser"},"date":"2023-09-28 09:20:31.000000000","message":"Patch Set 4: Published edit on patch set 3.\n\nOutdated Votes:\n* Verified-1\n","accounts_in_message":[],"_revision_number":4},{"id":"91257cd885ed294aa032b03d10c0638d6233328f","tag":"autogenerated:zuul:check","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2023-09-28 09:21:00.000000000","message":"Patch Set 4: Verified-1\n\nMerge Failed.\n\nThis change or one of its cross-repo dependencies was unable to be automatically merged with the current state of its repository. Please rebase the change and upload a new patchset.\nWarning:\n  Error merging gerrit/openstack/openstack-ansible-lxc_hosts for 894557,4","accounts_in_message":[],"_revision_number":4},{"id":"2650d8ad7ec5dfbfd2222bc5b5aeeaa6f46d934a","tag":"autogenerated:gerrit:newPatchSet","author":{"_account_id":25023,"name":"Jonathan Rosser","email":"jonathan.rosser@rd.bbc.co.uk","username":"jrosser"},"date":"2023-09-28 09:27:01.000000000","message":"Uploaded patch set 5.\n\nOutdated Votes:\n* Verified-1\n","accounts_in_message":[],"_revision_number":5},{"id":"4baf7617a1dee66bb16a6e2f2b73b2aedd0f96d6","tag":"autogenerated:zuul:check","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2023-09-28 11:52:07.000000000","message":"Patch Set 5: Verified-1\n\nBuild failed (check pipeline).  For information on how to proceed, see\nhttps://docs.opendev.org/opendev/infra-manual/latest/developers.html#automated-testing\n\nhttps://zuul.opendev.org/t/openstack/buildset/ab172c161b5945ec8c28a47d42d56201\n\n- openstack-ansible-linters-ubuntu-jammy https://zuul.opendev.org/t/openstack/build/40b3efd667e5451786eb252659c3211a : SUCCESS in 20m 22s\n- openstack-tox-docs https://zuul.opendev.org/t/openstack/build/5074baa0ceb94690ba4d389af28b3183 : SUCCESS in 4m 39s\n- openstack-ansible-deploy-infra_lxc-centos-9-stream https://zuul.opendev.org/t/openstack/build/6c2708840c11437399d3a4fb694dc5ef : SUCCESS in 1h 41m 42s (non-voting)\n- openstack-ansible-deploy-infra_lxc-rockylinux-9 https://zuul.opendev.org/t/openstack/build/8113a4f43a664322b0de723e16f8ce4d : SUCCESS in 1h 45m 30s\n- openstack-ansible-deploy-infra_lxc_tls-rockylinux-9 https://zuul.opendev.org/t/openstack/build/fc6e9b92093a4681b8f320235f6e2c17 : SUCCESS in 1h 38m 09s\n- openstack-ansible-deploy-infra_lxc-debian-bullseye https://zuul.opendev.org/t/openstack/build/37cfb7de6d574a22a418421799b9b9c1 : SUCCESS in 1h 05m 21s\n- openstack-ansible-deploy-infra_lxc-ubuntu-jammy https://zuul.opendev.org/t/openstack/build/bd03d1c56b7544899dad1bc2288b54dd : SUCCESS in 1h 17m 51s\n- openstack-ansible-deploy-infra_lxc_stepca-ubuntu-jammy https://zuul.opendev.org/t/openstack/build/5dcd796b5bfe4ec4ba5785ae45a8436f : SUCCESS in 1h 21m 44s\n- openstack-ansible-deploy-infra_lxc_tls-ubuntu-jammy https://zuul.opendev.org/t/openstack/build/a87daf64d26849dc900bff9911bf45a3 : SUCCESS in 1h 14m 11s\n- openstack-ansible-upgrade-infra_lxc-rockylinux-9 https://zuul.opendev.org/t/openstack/build/bdd67f6f92804da0afbfd9753afb1176 : SUCCESS in 2h 14m 27s\n- openstack-ansible-upgrade-infra_lxc-ubuntu-jammy https://zuul.opendev.org/t/openstack/build/7cf0a73f13074156bb0cb3c35532f4b6 : SUCCESS in 1h 42m 32s\n- openstack-ansible-deploy-hosts_lxc_zfs-ubuntu-jammy https://zuul.opendev.org/t/openstack/build/a62c4c7ef2394b46bebcfdeb26189919 : SUCCESS in 30m 05s\n- openstack-ansible-deploy-hosts_lxc_btrfs-ubuntu-jammy https://zuul.opendev.org/t/openstack/build/b9a8d4befd93479fb6d35eb2aeb63b85 : FAILURE in 29m 28s\n- openstack-ansible-deploy-hosts_lxc_overlayfs-ubuntu-jammy https://zuul.opendev.org/t/openstack/build/6adce59965684f9e9b194a984cd42d67 : SUCCESS in 28m 59s\n- openstack-ansible-deploy-hosts_lxc_btrfs-debian-bullseye https://zuul.opendev.org/t/openstack/build/11e8f520e3fd4a398d509851afd8bc15 : SUCCESS in 26m 59s\n- openstack-ansible-deploy-hosts_lxc_overlayfs-debian-bullseye https://zuul.opendev.org/t/openstack/build/394f460d36704465b7956f8c37ae60c2 : SUCCESS in 29m 04s\n- openstack-ansible-deploy-hosts_lxc_overlayfs-rockylinux-9 https://zuul.opendev.org/t/openstack/build/f774b55af8f3404fb65f457032a5b863 : SUCCESS in 32m 05s","accounts_in_message":[],"_revision_number":5},{"id":"be066fe6cb750e4e188fc4dc27447dba32eedb16","author":{"_account_id":25023,"name":"Jonathan Rosser","email":"jonathan.rosser@rd.bbc.co.uk","username":"jrosser"},"date":"2023-09-28 12:58:35.000000000","message":"Patch Set 5:\n\n(1 comment)","accounts_in_message":[],"_revision_number":5},{"id":"c25b33205a7ad026e59294038f82ec54060b9d8b","tag":"autogenerated:zuul:check","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2023-09-28 15:17:07.000000000","message":"Patch Set 5: Verified+1\n\nBuild succeeded (check pipeline).\nhttps://zuul.opendev.org/t/openstack/buildset/bbc852cdc89942188c82f00862e1bb90\n\n- openstack-ansible-linters-ubuntu-jammy https://zuul.opendev.org/t/openstack/build/9b41dea514824c1cba83a127241230f8 : SUCCESS in 11m 07s\n- openstack-tox-docs https://zuul.opendev.org/t/openstack/build/b3be862498394491983ca3301e48a208 : SUCCESS in 5m 18s\n- openstack-ansible-deploy-infra_lxc-centos-9-stream https://zuul.opendev.org/t/openstack/build/ca807f9950064e31a3f30b1006e930bc : SUCCESS in 1h 20m 50s (non-voting)\n- openstack-ansible-deploy-infra_lxc-rockylinux-9 https://zuul.opendev.org/t/openstack/build/4dfbe4d08fe2478b9d4f3e040b811381 : SUCCESS in 1h 40m 25s\n- openstack-ansible-deploy-infra_lxc_tls-rockylinux-9 https://zuul.opendev.org/t/openstack/build/649496b2edf74b01a7255359e472ba1f : SUCCESS in 1h 34m 41s\n- openstack-ansible-deploy-infra_lxc-debian-bullseye https://zuul.opendev.org/t/openstack/build/a70ca34543e549e6a3eadf350475044b : SUCCESS in 1h 11m 02s\n- openstack-ansible-deploy-infra_lxc-ubuntu-jammy https://zuul.opendev.org/t/openstack/build/83393b8cc9ad4471bdfc7803d87a9359 : SUCCESS in 1h 18m 59s\n- openstack-ansible-deploy-infra_lxc_stepca-ubuntu-jammy https://zuul.opendev.org/t/openstack/build/0dc5cfa9766740a1ae322faf337782d7 : SUCCESS in 1h 16m 26s\n- openstack-ansible-deploy-infra_lxc_tls-ubuntu-jammy https://zuul.opendev.org/t/openstack/build/afb6b5d474b94123ba61ce05938982c6 : SUCCESS in 1h 21m 10s\n- openstack-ansible-upgrade-infra_lxc-rockylinux-9 https://zuul.opendev.org/t/openstack/build/cb44194526604aae87f8d5083d3363e2 : SUCCESS in 2h 12m 10s\n- openstack-ansible-upgrade-infra_lxc-ubuntu-jammy https://zuul.opendev.org/t/openstack/build/56393e755df441e1b0d459cb277ee81b : SUCCESS in 1h 46m 58s\n- openstack-ansible-deploy-hosts_lxc_zfs-ubuntu-jammy https://zuul.opendev.org/t/openstack/build/84bd77169e7f419f937350b8a7d9c5f9 : SUCCESS in 29m 20s\n- openstack-ansible-deploy-hosts_lxc_btrfs-ubuntu-jammy https://zuul.opendev.org/t/openstack/build/899c5159f3c54e6f88e27a0f7e580cf8 : SUCCESS in 28m 56s\n- openstack-ansible-deploy-hosts_lxc_overlayfs-ubuntu-jammy https://zuul.opendev.org/t/openstack/build/540017734f284aa1912aec5ddf80d5dc : SUCCESS in 29m 29s\n- openstack-ansible-deploy-hosts_lxc_btrfs-debian-bullseye https://zuul.opendev.org/t/openstack/build/d00ff4d171694df58113f1ef8a47aaea : SUCCESS in 26m 32s\n- openstack-ansible-deploy-hosts_lxc_overlayfs-debian-bullseye https://zuul.opendev.org/t/openstack/build/e96df31c39fb45bf9d6c220f865276ed : SUCCESS in 26m 53s\n- openstack-ansible-deploy-hosts_lxc_overlayfs-rockylinux-9 https://zuul.opendev.org/t/openstack/build/144167804016414f91d93156b8cf690f : SUCCESS in 34m 22s","accounts_in_message":[],"_revision_number":5},{"id":"d813a73815dbbdf43677191d7a938b9d1a11654c","author":{"_account_id":28619,"name":"Dmitriy Rabotyagov","email":"noonedeadpunk@gmail.com","username":"noonedeadpunk"},"date":"2023-09-29 08:10:10.000000000","message":"Patch Set 5: Code-Review+2","accounts_in_message":[],"_revision_number":5},{"id":"1437dc8229426d58ceaba0b4db2be9b7dd9bf062","tag":"autogenerated:gerrit:newPatchSet","author":{"_account_id":25023,"name":"Jonathan Rosser","email":"jonathan.rosser@rd.bbc.co.uk","username":"jrosser"},"date":"2023-09-29 12:44:05.000000000","message":"Patch Set 6: Patch Set 5 was rebased\n\nCopied Votes:\n* Code-Review+2 (copy condition: \"changekind:TRIVIAL_REBASE OR is:MIN\")\n\nOutdated Votes:\n* Verified+1\n","accounts_in_message":[],"_revision_number":6},{"id":"51ce6b4fee2857a238e65555bf83b41598de8e2b","tag":"autogenerated:zuul:check","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2023-09-29 14:48:10.000000000","message":"Patch Set 6: Verified-1\n\nBuild failed (check pipeline).  For information on how to proceed, see\nhttps://docs.opendev.org/opendev/infra-manual/latest/developers.html#automated-testing\n\nhttps://zuul.opendev.org/t/openstack/buildset/1fe79f88e99d49b28d954061b8e765d7\n\n- openstack-ansible-linters-ubuntu-jammy https://zuul.opendev.org/t/openstack/build/849b7c1aa12b4b12a16595ee1a556c0b : SUCCESS in 10m 51s\n- openstack-tox-docs https://zuul.opendev.org/t/openstack/build/9cf93724f8df4497b4414b091dca90da : SUCCESS in 4m 40s\n- openstack-ansible-deploy-infra_lxc-centos-9-stream https://zuul.opendev.org/t/openstack/build/3a1f5f284a31493cbcc03f9de9073f81 : SUCCESS in 1h 29m 28s (non-voting)\n- openstack-ansible-deploy-infra_lxc-rockylinux-9 https://zuul.opendev.org/t/openstack/build/895c8cd2b51844cc86dc82793b0fbb47 : SUCCESS in 1h 38m 26s\n- openstack-ansible-deploy-infra_lxc_tls-rockylinux-9 https://zuul.opendev.org/t/openstack/build/da75bf42bf664870becd810c711dbac0 : SUCCESS in 1h 22m 08s\n- openstack-ansible-deploy-infra_lxc-debian-bullseye https://zuul.opendev.org/t/openstack/build/39a4647a80274119917a79e59effac45 : SUCCESS in 1h 18m 20s\n- openstack-ansible-deploy-infra_lxc-ubuntu-jammy https://zuul.opendev.org/t/openstack/build/c57cd18213204f568f3dfc826d750ad3 : SUCCESS in 1h 29m 58s\n- openstack-ansible-deploy-infra_lxc_stepca-ubuntu-jammy https://zuul.opendev.org/t/openstack/build/7d5c3b6a7d9e4e8ab217a8510a9c54fe : SUCCESS in 1h 24m 59s\n- openstack-ansible-deploy-infra_lxc_tls-ubuntu-jammy https://zuul.opendev.org/t/openstack/build/d099740724144790bc7b692f681d81c1 : SUCCESS in 1h 26m 28s\n- openstack-ansible-upgrade-infra_lxc-rockylinux-9 https://zuul.opendev.org/t/openstack/build/aae9c0f97f6a47a0bbf6d782abdf2131 : SUCCESS in 1h 54m 30s\n- openstack-ansible-upgrade-infra_lxc-ubuntu-jammy https://zuul.opendev.org/t/openstack/build/27764ee4217344c68219c1702e134511 : SUCCESS in 1h 43m 14s\n- openstack-ansible-deploy-hosts_lxc_zfs-ubuntu-jammy https://zuul.opendev.org/t/openstack/build/56881d4257984cea9c77846518f6fc7e : SUCCESS in 30m 38s\n- openstack-ansible-deploy-hosts_lxc_btrfs-ubuntu-jammy https://zuul.opendev.org/t/openstack/build/692f85c2ba6441478e86125293496c44 : SUCCESS in 27m 26s\n- openstack-ansible-deploy-hosts_lxc_overlayfs-ubuntu-jammy https://zuul.opendev.org/t/openstack/build/e02623c9fde14773b318196570d52517 : FAILURE in 34m 51s\n- openstack-ansible-deploy-hosts_lxc_btrfs-debian-bullseye https://zuul.opendev.org/t/openstack/build/94112c8dde1e477ebf1e91e6d4026430 : SUCCESS in 24m 52s\n- openstack-ansible-deploy-hosts_lxc_overlayfs-debian-bullseye https://zuul.opendev.org/t/openstack/build/e1646bf04b604bda8e6fe69bb21807b4 : SUCCESS in 32m 29s\n- openstack-ansible-deploy-hosts_lxc_overlayfs-rockylinux-9 https://zuul.opendev.org/t/openstack/build/783d3b2a458945c285a033064153af89 : SUCCESS in 36m 35s","accounts_in_message":[],"_revision_number":6},{"id":"6a6ee66b2298ec5b525cef196cfd2bdd5d321a6f","author":{"_account_id":28619,"name":"Dmitriy Rabotyagov","email":"noonedeadpunk@gmail.com","username":"noonedeadpunk"},"date":"2023-10-02 06:50:46.000000000","message":"Patch Set 6: Code-Review+2\n\n(1 comment)","accounts_in_message":[],"_revision_number":6},{"id":"dd9747f1e4481ab95b2a5c1defb31c552ea9773e","tag":"autogenerated:zuul:check","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2023-10-02 09:10:14.000000000","message":"Patch Set 6:\n\nBuild failed (check pipeline).  For information on how to proceed, see\nhttps://docs.opendev.org/opendev/infra-manual/latest/developers.html#automated-testing\n\nhttps://zuul.opendev.org/t/openstack/buildset/3cedbb8956f94a4fb68ad5f47a06adf2\n\n- openstack-ansible-linters-ubuntu-jammy https://zuul.opendev.org/t/openstack/build/a6b96e68061f4ae196bd48111082d710 : SUCCESS in 11m 49s\n- openstack-tox-docs https://zuul.opendev.org/t/openstack/build/51dd1d31b1ec4c93b0bb834e263bb560 : SUCCESS in 5m 03s\n- openstack-ansible-deploy-infra_lxc-centos-9-stream https://zuul.opendev.org/t/openstack/build/d84ac0db998f48518e1cd505647e7fc2 : SUCCESS in 1h 35m 09s (non-voting)\n- openstack-ansible-deploy-infra_lxc-rockylinux-9 https://zuul.opendev.org/t/openstack/build/23937b3155054af88555c55cc12a9b75 : SUCCESS in 1h 37m 42s\n- openstack-ansible-deploy-infra_lxc_tls-rockylinux-9 https://zuul.opendev.org/t/openstack/build/7c6e8d9986d546c68243741e5dc2f8dc : SUCCESS in 1h 36m 40s\n- openstack-ansible-deploy-infra_lxc-debian-bullseye https://zuul.opendev.org/t/openstack/build/15906e245af64558b02f38170e4a3559 : SUCCESS in 1h 16m 33s\n- openstack-ansible-deploy-infra_lxc-ubuntu-jammy https://zuul.opendev.org/t/openstack/build/2aa2722aa9be481b8c2a03dae9cf5b4a : SUCCESS in 1h 19m 23s\n- openstack-ansible-deploy-infra_lxc_stepca-ubuntu-jammy https://zuul.opendev.org/t/openstack/build/46277226a03f489483a47d0a02dd4fc7 : SUCCESS in 1h 19m 45s\n- openstack-ansible-deploy-infra_lxc_tls-ubuntu-jammy https://zuul.opendev.org/t/openstack/build/782d03b7bb354fcbbefa9f542965587d : SUCCESS in 1h 24m 17s\n- openstack-ansible-upgrade-infra_lxc-rockylinux-9 https://zuul.opendev.org/t/openstack/build/c05d80032d06420b816699eba1d01f57 : SUCCESS in 2h 10m 55s\n- openstack-ansible-upgrade-infra_lxc-ubuntu-jammy https://zuul.opendev.org/t/openstack/build/8fa701d694a4466ab2bd54227594e2b6 : SUCCESS in 1h 27m 48s\n- openstack-ansible-deploy-hosts_lxc_zfs-ubuntu-jammy https://zuul.opendev.org/t/openstack/build/1b4d64b6b0944e828028dc772d92fc0f : SUCCESS in 23m 55s\n- openstack-ansible-deploy-hosts_lxc_btrfs-ubuntu-jammy https://zuul.opendev.org/t/openstack/build/87435871deff4691a9a6e4dc389731ec : SUCCESS in 23m 06s\n- openstack-ansible-deploy-hosts_lxc_overlayfs-ubuntu-jammy https://zuul.opendev.org/t/openstack/build/c58a3f4df3074aeaaf8220ba3984482a : SUCCESS in 24m 24s\n- openstack-ansible-deploy-hosts_lxc_btrfs-debian-bullseye https://zuul.opendev.org/t/openstack/build/ec53014481694c04bef0ffcee0111d51 : FAILURE in 24m 42s\n- openstack-ansible-deploy-hosts_lxc_overlayfs-debian-bullseye https://zuul.opendev.org/t/openstack/build/810ca47c8f5741779fe52ad4d58e155d : FAILURE in 22m 43s\n- openstack-ansible-deploy-hosts_lxc_overlayfs-rockylinux-9 https://zuul.opendev.org/t/openstack/build/ea3ea53f680447b7981da0046e418e68 : SUCCESS in 29m 14s","accounts_in_message":[],"_revision_number":6},{"id":"504fcf21038dfc77215552bb7fbdcc7857182404","author":{"_account_id":28619,"name":"Dmitriy Rabotyagov","email":"noonedeadpunk@gmail.com","username":"noonedeadpunk"},"date":"2023-10-04 16:23:25.000000000","message":"Patch Set 6:\n\n(1 comment)","accounts_in_message":[],"_revision_number":6},{"id":"66b41f55047c278fd91bb9f94f8f34e214a48c0e","tag":"autogenerated:zuul:check","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2023-10-04 18:58:12.000000000","message":"Patch Set 6: Verified+1\n\nBuild succeeded (check pipeline).\nhttps://zuul.opendev.org/t/openstack/buildset/2ac31162b12f4ac5a6201d13f79540bb\n\n- openstack-ansible-linters-ubuntu-jammy https://zuul.opendev.org/t/openstack/build/041ec930ae6d40188a4032eacfa8cae4 : SUCCESS in 20m 14s\n- openstack-tox-docs https://zuul.opendev.org/t/openstack/build/e4e98f3edb8d42a19a7ac9b4007ce29e : SUCCESS in 7m 37s\n- openstack-ansible-deploy-infra_lxc-centos-9-stream https://zuul.opendev.org/t/openstack/build/836874a9a832442085f0bfd1053df4c0 : SUCCESS in 1h 32m 58s (non-voting)\n- openstack-ansible-deploy-infra_lxc-rockylinux-9 https://zuul.opendev.org/t/openstack/build/fb885cc5705946eabbc3d0f290992b1f : SUCCESS in 1h 38m 56s\n- openstack-ansible-deploy-infra_lxc_tls-rockylinux-9 https://zuul.opendev.org/t/openstack/build/02aea8e422ba43169f98ddff2a729cd3 : SUCCESS in 1h 44m 06s\n- openstack-ansible-deploy-infra_lxc-debian-bullseye https://zuul.opendev.org/t/openstack/build/7511635bd94148cd8663b55832451fe1 : SUCCESS in 1h 23m 06s\n- openstack-ansible-deploy-infra_lxc-ubuntu-jammy https://zuul.opendev.org/t/openstack/build/2d3366e8b75145d4b57979872d979f76 : SUCCESS in 1h 12m 42s\n- openstack-ansible-deploy-infra_lxc_stepca-ubuntu-jammy https://zuul.opendev.org/t/openstack/build/85cb8104332243728dd06eada9b9174b : SUCCESS in 1h 26m 38s\n- openstack-ansible-deploy-infra_lxc_tls-ubuntu-jammy https://zuul.opendev.org/t/openstack/build/85367d1a6bb64514a24341a649da0bce : SUCCESS in 1h 29m 17s\n- openstack-ansible-upgrade-infra_lxc-rockylinux-9 https://zuul.opendev.org/t/openstack/build/98dbcc7f49c74d44827de1bdfa58a55f : SUCCESS in 2h 22m 01s\n- openstack-ansible-upgrade-infra_lxc-ubuntu-jammy https://zuul.opendev.org/t/openstack/build/6185278ffc7d4238b59f23277a33c997 : SUCCESS in 1h 32m 40s\n- openstack-ansible-deploy-hosts_lxc_zfs-ubuntu-jammy https://zuul.opendev.org/t/openstack/build/1b4780aac2434dc08f315a2f4badaeb9 : SUCCESS in 40m 08s\n- openstack-ansible-deploy-hosts_lxc_btrfs-ubuntu-jammy https://zuul.opendev.org/t/openstack/build/3e279b431457480cb3068e2a58bcd502 : SUCCESS in 31m 39s\n- openstack-ansible-deploy-hosts_lxc_overlayfs-ubuntu-jammy https://zuul.opendev.org/t/openstack/build/eee39a3b87034d4b90c73302cf1cb1e4 : SUCCESS in 39m 05s\n- openstack-ansible-deploy-hosts_lxc_btrfs-debian-bullseye https://zuul.opendev.org/t/openstack/build/e9a9a35e633c44229c2ca608b7854c74 : SUCCESS in 33m 37s\n- openstack-ansible-deploy-hosts_lxc_overlayfs-debian-bullseye https://zuul.opendev.org/t/openstack/build/81e85cfaf3d54593972aad0990a5e3ee : SUCCESS in 33m 17s\n- openstack-ansible-deploy-hosts_lxc_overlayfs-rockylinux-9 https://zuul.opendev.org/t/openstack/build/65441f4d96f74976b19416154d1118c7 : SUCCESS in 30m 50s","accounts_in_message":[],"_revision_number":6},{"id":"1215c59a4a0f9d62c3638bd6c774ca55c1a99791","author":{"_account_id":31542,"name":"Andrew Bonney","email":"andrew.bonney@bbc.co.uk","username":"andrewbonney"},"date":"2023-10-10 08:43:21.000000000","message":"Patch Set 6: Code-Review+2 Workflow+1","accounts_in_message":[],"_revision_number":6},{"id":"97c7fee3df3a4f14e7a04398f29d34bea485a278","tag":"autogenerated:zuul:gate","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2023-10-10 08:45:30.000000000","message":"Patch Set 6: -Verified\n\nStarting gate jobs.","accounts_in_message":[],"_revision_number":6},{"id":"e806f5cbe8994d12280dfc15b94f58919868cc7a","tag":"autogenerated:zuul:gate","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2023-10-10 11:16:38.000000000","message":"Patch Set 6: Verified+2\n\nBuild succeeded (gate pipeline).\nhttps://zuul.opendev.org/t/openstack/buildset/f7acb0dc7e104ec38451728eb1840eb4\n\n- openstack-ansible-linters-ubuntu-jammy https://zuul.opendev.org/t/openstack/build/c3b3e88d1ee949a9bd58534fb1952b94 : SUCCESS in 18m 00s\n- openstack-tox-docs https://zuul.opendev.org/t/openstack/build/16ea8e8ffa514fb994beb002b19ff924 : SUCCESS in 6m 14s\n- openstack-ansible-deploy-infra_lxc-rockylinux-9 https://zuul.opendev.org/t/openstack/build/5a7a0cddf7ed4cdebbc5e5314a2feb13 : SUCCESS in 1h 31m 22s\n- openstack-ansible-deploy-infra_lxc_tls-rockylinux-9 https://zuul.opendev.org/t/openstack/build/e8c608d961fc4d95b4e5a56da4ac005b : SUCCESS in 1h 44m 47s\n- openstack-ansible-deploy-infra_lxc-debian-bullseye https://zuul.opendev.org/t/openstack/build/69efe1157235463381fd5bc3ff71c5d8 : SUCCESS in 1h 10m 10s\n- openstack-ansible-deploy-infra_lxc-ubuntu-jammy https://zuul.opendev.org/t/openstack/build/004f019a917e4ab5a6d84a73b4f3a9bf : SUCCESS in 1h 10m 06s\n- openstack-ansible-deploy-infra_lxc_stepca-ubuntu-jammy https://zuul.opendev.org/t/openstack/build/fff808329abf47ffb97778ae02b9d3a7 : SUCCESS in 1h 54m 59s\n- openstack-ansible-deploy-infra_lxc_tls-ubuntu-jammy https://zuul.opendev.org/t/openstack/build/0a9a2ad6359549ec9dc87c91c2881c67 : SUCCESS in 1h 17m 55s\n- openstack-ansible-upgrade-infra_lxc-rockylinux-9 https://zuul.opendev.org/t/openstack/build/3ce9f57653db48bc9448093d93839fac : SUCCESS in 2h 03m 38s\n- openstack-ansible-upgrade-infra_lxc-ubuntu-jammy https://zuul.opendev.org/t/openstack/build/5260a4013f8a4cab9a08398be646ebc7 : SUCCESS in 1h 34m 19s\n- openstack-ansible-deploy-hosts_lxc_zfs-ubuntu-jammy https://zuul.opendev.org/t/openstack/build/74fb614c069d40b1b7813d8932bbedde : SUCCESS in 27m 18s\n- openstack-ansible-deploy-hosts_lxc_btrfs-ubuntu-jammy https://zuul.opendev.org/t/openstack/build/d001d8ccd3014aaf9e61c052651231b4 : SUCCESS in 26m 20s\n- openstack-ansible-deploy-hosts_lxc_overlayfs-ubuntu-jammy https://zuul.opendev.org/t/openstack/build/166b93d0ef42492194c06c89110809c5 : SUCCESS in 29m 06s\n- openstack-ansible-deploy-hosts_lxc_btrfs-debian-bullseye https://zuul.opendev.org/t/openstack/build/3e7e1c9041cd4b6db6fe91b8e99e8d53 : SUCCESS in 24m 51s\n- openstack-ansible-deploy-hosts_lxc_overlayfs-debian-bullseye https://zuul.opendev.org/t/openstack/build/b09b56e375a64e39a603214716dddb9a : SUCCESS in 29m 00s\n- openstack-ansible-deploy-hosts_lxc_overlayfs-rockylinux-9 https://zuul.opendev.org/t/openstack/build/f599c3f9f82542a9b6906ffbc36ed716 : SUCCESS in 30m 02s","accounts_in_message":[],"_revision_number":6},{"id":"786666c80f9c02017dc0f2df5701f3af83c1e3bb","tag":"autogenerated:gerrit:merged","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2023-10-10 11:16:38.000000000","message":"Change has been successfully merged","accounts_in_message":[],"_revision_number":6},{"id":"b925a34528c7a47edfd3e9d4c0d55f7a44f7ffee","tag":"autogenerated:zuul:promote","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2023-10-10 11:17:44.000000000","message":"Patch Set 6:\n\nBuild succeeded (promote pipeline).\nhttps://zuul.opendev.org/t/openstack/buildset/a1f5cef018a94d14ae9142a1c98e8ba1\n\n- promote-openstack-tox-docs https://zuul.opendev.org/t/openstack/build/8acb8394bdac40ce8073c829d3509aba : SUCCESS in 43s","accounts_in_message":[],"_revision_number":6}],"current_revision_number":6,"current_revision":"03dc91fa85d3c717ec0de00ba894999f62234193","revisions":{"5d79bec6a175acfa37c960c15235d13fbe7147c6":{"kind":"REWORK","_number":1,"created":"2023-09-11 16:14:19.000000000","uploader":{"_account_id":25023,"name":"Jonathan Rosser","email":"jonathan.rosser@rd.bbc.co.uk","username":"jrosser"},"ref":"refs/changes/57/894557/1","fetch":{"anonymous http":{"url":"https://review.opendev.org/openstack/openstack-ansible-lxc_hosts","ref":"refs/changes/57/894557/1","commands":{"Checkout":"git fetch https://review.opendev.org/openstack/openstack-ansible-lxc_hosts refs/changes/57/894557/1 \u0026\u0026 git checkout FETCH_HEAD","Cherry Pick":"git fetch https://review.opendev.org/openstack/openstack-ansible-lxc_hosts refs/changes/57/894557/1 \u0026\u0026 git cherry-pick FETCH_HEAD","Format Patch":"git fetch https://review.opendev.org/openstack/openstack-ansible-lxc_hosts refs/changes/57/894557/1 \u0026\u0026 git format-patch -1 --stdout FETCH_HEAD","Pull":"git pull https://review.opendev.org/openstack/openstack-ansible-lxc_hosts refs/changes/57/894557/1"}}},"commit":{"parents":[{"commit":"4a8bc6de280b5f0839abb3fb2aa4516f6c0bc6e3","subject":"Sync additional apt config from the host to the container base image","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/openstack-ansible-lxc_hosts/commit/4a8bc6de280b5f0839abb3fb2aa4516f6c0bc6e3"}]}],"author":{"name":"Jonathan Rosser","email":"jonathan.rosser@rd.bbc.co.uk","date":"2023-09-11 16:03:34.000000000","tz":60},"committer":{"name":"Jonathan Rosser","email":"jonathan.rosser@rd.bbc.co.uk","date":"2023-09-11 16:13:27.000000000","tz":60},"subject":"Add ca-certificates into debian base image during debootstrap.","message":"Add ca-certificates into debian base image during debootstrap.\n\ndebootstrap uses http for it\u0027s apt config so can function without\nthe certificates from ca-certificates being installed.\n\nThe debian bookworm cloud image defaults to using https for the\napt repo urls, so unless the ca-certificates package is present\nno more apt operations can be done once the apt configuration is\nsynchronised from the host to the container image.\n\nInstalling ca-certificates during the initial debootstrap avoids\nthe issue of not being able to install ca-certificates due to failed\nSSL verification.\n\nChange-Id: Ia78429eaf4bd71a8f3509c4e484f7dd02574c6b1\n","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/openstack-ansible-lxc_hosts/commit/5d79bec6a175acfa37c960c15235d13fbe7147c6"}],"resolve_conflicts_web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/openstack-ansible-lxc_hosts/commit/5d79bec6a175acfa37c960c15235d13fbe7147c6"}]},"branch":"refs/heads/master"},"cce5613c75113781e59082be401249214a37ccaf":{"kind":"TRIVIAL_REBASE","_number":2,"created":"2023-09-11 18:30:02.000000000","uploader":{"_account_id":25023,"name":"Jonathan Rosser","email":"jonathan.rosser@rd.bbc.co.uk","username":"jrosser"},"ref":"refs/changes/57/894557/2","fetch":{"anonymous http":{"url":"https://review.opendev.org/openstack/openstack-ansible-lxc_hosts","ref":"refs/changes/57/894557/2","commands":{"Checkout":"git fetch https://review.opendev.org/openstack/openstack-ansible-lxc_hosts refs/changes/57/894557/2 \u0026\u0026 git checkout FETCH_HEAD","Cherry Pick":"git fetch https://review.opendev.org/openstack/openstack-ansible-lxc_hosts refs/changes/57/894557/2 \u0026\u0026 git cherry-pick FETCH_HEAD","Format Patch":"git fetch https://review.opendev.org/openstack/openstack-ansible-lxc_hosts refs/changes/57/894557/2 \u0026\u0026 git format-patch -1 --stdout FETCH_HEAD","Pull":"git pull https://review.opendev.org/openstack/openstack-ansible-lxc_hosts refs/changes/57/894557/2"}}},"commit":{"parents":[{"commit":"02297473ef32c3a49b74e676889f039da920c56d","subject":"Sync additional apt config from the host to the container base image","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/openstack-ansible-lxc_hosts/commit/02297473ef32c3a49b74e676889f039da920c56d"}]}],"author":{"name":"Jonathan Rosser","email":"jonathan.rosser@rd.bbc.co.uk","date":"2023-09-11 16:03:34.000000000","tz":60},"committer":{"name":"Jonathan Rosser","email":"jonathan.rosser@rd.bbc.co.uk","date":"2023-09-11 18:29:34.000000000","tz":60},"subject":"Add ca-certificates into debian base image during debootstrap.","message":"Add ca-certificates into debian base image during debootstrap.\n\ndebootstrap uses http for it\u0027s apt config so can function without\nthe certificates from ca-certificates being installed.\n\nThe debian bookworm cloud image defaults to using https for the\napt repo urls, so unless the ca-certificates package is present\nno more apt operations can be done once the apt configuration is\nsynchronised from the host to the container image.\n\nInstalling ca-certificates during the initial debootstrap avoids\nthe issue of not being able to install ca-certificates due to failed\nSSL verification.\n\nChange-Id: Ia78429eaf4bd71a8f3509c4e484f7dd02574c6b1\n","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/openstack-ansible-lxc_hosts/commit/cce5613c75113781e59082be401249214a37ccaf"}],"resolve_conflicts_web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/openstack-ansible-lxc_hosts/commit/cce5613c75113781e59082be401249214a37ccaf"}]},"branch":"refs/heads/master"},"c4e272074966c00052cfe74545000ecca9303b18":{"kind":"TRIVIAL_REBASE","_number":3,"created":"2023-09-27 16:29:37.000000000","uploader":{"_account_id":25023,"name":"Jonathan Rosser","email":"jonathan.rosser@rd.bbc.co.uk","username":"jrosser"},"ref":"refs/changes/57/894557/3","fetch":{"anonymous http":{"url":"https://review.opendev.org/openstack/openstack-ansible-lxc_hosts","ref":"refs/changes/57/894557/3","commands":{"Checkout":"git fetch https://review.opendev.org/openstack/openstack-ansible-lxc_hosts refs/changes/57/894557/3 \u0026\u0026 git checkout FETCH_HEAD","Cherry Pick":"git fetch https://review.opendev.org/openstack/openstack-ansible-lxc_hosts refs/changes/57/894557/3 \u0026\u0026 git cherry-pick FETCH_HEAD","Format Patch":"git fetch https://review.opendev.org/openstack/openstack-ansible-lxc_hosts refs/changes/57/894557/3 \u0026\u0026 git format-patch -1 --stdout FETCH_HEAD","Pull":"git pull https://review.opendev.org/openstack/openstack-ansible-lxc_hosts refs/changes/57/894557/3"}}},"commit":{"parents":[{"commit":"f435f144aa58ec73ef7b21925f435e03f435b300","subject":"Sync additional apt config from the host to the container base image","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/openstack-ansible-lxc_hosts/commit/f435f144aa58ec73ef7b21925f435e03f435b300"}]}],"author":{"name":"Jonathan Rosser","email":"jonathan.rosser@rd.bbc.co.uk","date":"2023-09-11 16:03:34.000000000","tz":60},"committer":{"name":"Jonathan Rosser","email":"jonathan.rosser@rd.bbc.co.uk","date":"2023-09-27 16:29:37.000000000","tz":0},"subject":"Add ca-certificates into debian base image during debootstrap.","message":"Add ca-certificates into debian base image during debootstrap.\n\ndebootstrap uses http for it\u0027s apt config so can function without\nthe certificates from ca-certificates being installed.\n\nThe debian bookworm cloud image defaults to using https for the\napt repo urls, so unless the ca-certificates package is present\nno more apt operations can be done once the apt configuration is\nsynchronised from the host to the container image.\n\nInstalling ca-certificates during the initial debootstrap avoids\nthe issue of not being able to install ca-certificates due to failed\nSSL verification.\n\nChange-Id: Ia78429eaf4bd71a8f3509c4e484f7dd02574c6b1\n","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/openstack-ansible-lxc_hosts/commit/c4e272074966c00052cfe74545000ecca9303b18"}],"resolve_conflicts_web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/openstack-ansible-lxc_hosts/commit/c4e272074966c00052cfe74545000ecca9303b18"}]},"branch":"refs/heads/master","description":"Rebase"},"f1b02d0989f03e5b2380f76cfde695dcf92dff47":{"kind":"REWORK","_number":4,"created":"2023-09-28 09:20:31.000000000","uploader":{"_account_id":25023,"name":"Jonathan Rosser","email":"jonathan.rosser@rd.bbc.co.uk","username":"jrosser"},"ref":"refs/changes/57/894557/4","fetch":{"anonymous http":{"url":"https://review.opendev.org/openstack/openstack-ansible-lxc_hosts","ref":"refs/changes/57/894557/4","commands":{"Checkout":"git fetch https://review.opendev.org/openstack/openstack-ansible-lxc_hosts refs/changes/57/894557/4 \u0026\u0026 git checkout FETCH_HEAD","Cherry Pick":"git fetch https://review.opendev.org/openstack/openstack-ansible-lxc_hosts refs/changes/57/894557/4 \u0026\u0026 git cherry-pick FETCH_HEAD","Format Patch":"git fetch https://review.opendev.org/openstack/openstack-ansible-lxc_hosts refs/changes/57/894557/4 \u0026\u0026 git format-patch -1 --stdout FETCH_HEAD","Pull":"git pull https://review.opendev.org/openstack/openstack-ansible-lxc_hosts refs/changes/57/894557/4"}}},"commit":{"parents":[{"commit":"f435f144aa58ec73ef7b21925f435e03f435b300","subject":"Sync additional apt config from the host to the container base image","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/openstack-ansible-lxc_hosts/commit/f435f144aa58ec73ef7b21925f435e03f435b300"}]}],"author":{"name":"Jonathan Rosser","email":"jonathan.rosser@rd.bbc.co.uk","date":"2023-09-11 16:03:34.000000000","tz":60},"committer":{"name":"Jonathan Rosser","email":"jonathan.rosser@rd.bbc.co.uk","date":"2023-09-28 09:20:31.000000000","tz":0},"subject":"Add ca-certificates into debian base image during debootstrap.","message":"Add ca-certificates into debian base image during debootstrap.\n\ndebootstrap uses http for it\u0027s apt config so can function without\nthe certificates from ca-certificates being installed.\n\nThe debian bookworm cloud image defaults to using https for the\napt repo urls, so unless the ca-certificates package is present\nno more apt operations can be done once the apt configuration is\nsynchronised from the host to the container image.\n\nInstalling ca-certificates during the initial debootstrap avoids\nthe issue of not being able to install ca-certificates due to failed\nSSL verification.\n\nChange-Id: Ia78429eaf4bd71a8f3509c4e484f7dd02574c6b1\n","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/openstack-ansible-lxc_hosts/commit/f1b02d0989f03e5b2380f76cfde695dcf92dff47"}],"resolve_conflicts_web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/openstack-ansible-lxc_hosts/commit/f1b02d0989f03e5b2380f76cfde695dcf92dff47"}]},"branch":"refs/heads/master"},"efbbcedc1d26b435552e02e8e355fb9b0e53e916":{"kind":"REWORK","_number":5,"created":"2023-09-28 09:27:01.000000000","uploader":{"_account_id":25023,"name":"Jonathan Rosser","email":"jonathan.rosser@rd.bbc.co.uk","username":"jrosser"},"ref":"refs/changes/57/894557/5","fetch":{"anonymous http":{"url":"https://review.opendev.org/openstack/openstack-ansible-lxc_hosts","ref":"refs/changes/57/894557/5","commands":{"Checkout":"git fetch https://review.opendev.org/openstack/openstack-ansible-lxc_hosts refs/changes/57/894557/5 \u0026\u0026 git checkout FETCH_HEAD","Cherry Pick":"git fetch https://review.opendev.org/openstack/openstack-ansible-lxc_hosts refs/changes/57/894557/5 \u0026\u0026 git cherry-pick FETCH_HEAD","Format Patch":"git fetch https://review.opendev.org/openstack/openstack-ansible-lxc_hosts refs/changes/57/894557/5 \u0026\u0026 git format-patch -1 --stdout FETCH_HEAD","Pull":"git pull https://review.opendev.org/openstack/openstack-ansible-lxc_hosts refs/changes/57/894557/5"}}},"commit":{"parents":[{"commit":"b6390ed2d3c4b4a7684957f93a30f1b5a49be420","subject":"Sync additional apt config from the host to the container base image","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/openstack-ansible-lxc_hosts/commit/b6390ed2d3c4b4a7684957f93a30f1b5a49be420"}]}],"author":{"name":"Jonathan Rosser","email":"jonathan.rosser@rd.bbc.co.uk","date":"2023-09-11 16:03:34.000000000","tz":60},"committer":{"name":"Jonathan Rosser","email":"jonathan.rosser@rd.bbc.co.uk","date":"2023-09-28 09:26:23.000000000","tz":60},"subject":"Add ca-certificates into debian base image during debootstrap.","message":"Add ca-certificates into debian base image during debootstrap.\n\ndebootstrap uses http for it\u0027s apt config so can function without\nthe certificates from ca-certificates being installed.\n\nThe debian bookworm cloud image defaults to using https for the\napt repo urls, so unless the ca-certificates package is present\nno more apt operations can be done once the apt configuration is\nsynchronised from the host to the container image.\n\nInstalling ca-certificates during the initial debootstrap avoids\nthe issue of not being able to install ca-certificates due to failed\nSSL verification.\n\nChange-Id: Ia78429eaf4bd71a8f3509c4e484f7dd02574c6b1\n","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/openstack-ansible-lxc_hosts/commit/efbbcedc1d26b435552e02e8e355fb9b0e53e916"}],"resolve_conflicts_web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/openstack-ansible-lxc_hosts/commit/efbbcedc1d26b435552e02e8e355fb9b0e53e916"}]},"branch":"refs/heads/master"},"03dc91fa85d3c717ec0de00ba894999f62234193":{"kind":"TRIVIAL_REBASE","_number":6,"created":"2023-09-29 12:44:05.000000000","uploader":{"_account_id":25023,"name":"Jonathan Rosser","email":"jonathan.rosser@rd.bbc.co.uk","username":"jrosser"},"ref":"refs/changes/57/894557/6","fetch":{"anonymous http":{"url":"https://review.opendev.org/openstack/openstack-ansible-lxc_hosts","ref":"refs/changes/57/894557/6","commands":{"Checkout":"git fetch https://review.opendev.org/openstack/openstack-ansible-lxc_hosts refs/changes/57/894557/6 \u0026\u0026 git checkout FETCH_HEAD","Cherry Pick":"git fetch https://review.opendev.org/openstack/openstack-ansible-lxc_hosts refs/changes/57/894557/6 \u0026\u0026 git cherry-pick FETCH_HEAD","Format Patch":"git fetch https://review.opendev.org/openstack/openstack-ansible-lxc_hosts refs/changes/57/894557/6 \u0026\u0026 git format-patch -1 --stdout FETCH_HEAD","Pull":"git pull https://review.opendev.org/openstack/openstack-ansible-lxc_hosts refs/changes/57/894557/6"}}},"commit":{"parents":[{"commit":"2506f0080a34cf48069954e26dc5a2fab23a310f","subject":"Sync additional apt config from the host to the container base image","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/openstack-ansible-lxc_hosts/commit/2506f0080a34cf48069954e26dc5a2fab23a310f"}]}],"author":{"name":"Jonathan Rosser","email":"jonathan.rosser@rd.bbc.co.uk","date":"2023-09-11 16:03:34.000000000","tz":60},"committer":{"name":"Jonathan Rosser","email":"jonathan.rosser@rd.bbc.co.uk","date":"2023-09-29 12:44:05.000000000","tz":0},"subject":"Add ca-certificates into debian base image during debootstrap.","message":"Add ca-certificates into debian base image during debootstrap.\n\ndebootstrap uses http for it\u0027s apt config so can function without\nthe certificates from ca-certificates being installed.\n\nThe debian bookworm cloud image defaults to using https for the\napt repo urls, so unless the ca-certificates package is present\nno more apt operations can be done once the apt configuration is\nsynchronised from the host to the container image.\n\nInstalling ca-certificates during the initial debootstrap avoids\nthe issue of not being able to install ca-certificates due to failed\nSSL verification.\n\nChange-Id: Ia78429eaf4bd71a8f3509c4e484f7dd02574c6b1\n","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/openstack-ansible-lxc_hosts/commit/03dc91fa85d3c717ec0de00ba894999f62234193"}],"resolve_conflicts_web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/openstack-ansible-lxc_hosts/commit/03dc91fa85d3c717ec0de00ba894999f62234193"}]},"branch":"refs/heads/master","description":"Rebase"}},"requirements":[],"submit_records":[{"rule_name":"gerrit~DefaultSubmitRule","status":"CLOSED","labels":[{"label":"Verified","status":"MAY","applied_by":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]}},{"label":"Code-Review","status":"MAY","applied_by":{"_account_id":31542,"name":"Andrew Bonney","email":"andrew.bonney@bbc.co.uk","username":"andrewbonney"}},{"label":"Workflow","status":"MAY","applied_by":{"_account_id":31542,"name":"Andrew Bonney","email":"andrew.bonney@bbc.co.uk","username":"andrewbonney"}},{"label":"Backport-Candidate","status":"MAY"}]}],"submit_requirements":[{"name":"Verified","description":"Verified in gate by CI","status":"SATISFIED","is_legacy":false,"submittability_expression_result":{"expression":"label:Verified\u003dMAX AND -label:Verified\u003dMIN","fulfilled":true,"status":"PASS","passing_atoms":["label:Verified\u003dMAX"],"failing_atoms":["label:Verified\u003dMIN"],"atom_explanations":{}}},{"name":"Backport-Candidate","description":"Backport candidate status","status":"NOT_APPLICABLE","is_legacy":false,"applicability_expression_result":{"fulfilled":false,"status":"FAIL"},"submittability_expression_result":{"expression":"is:true","fulfilled":true,"status":"NOT_EVALUATED","passing_atoms":[],"failing_atoms":[],"atom_explanations":{}}},{"name":"Code-Review","description":"Code reviewed by core reviewer","status":"SATISFIED","is_legacy":false,"submittability_expression_result":{"expression":"label:Code-Review\u003dMAX AND -label:Code-Review\u003dMIN","fulfilled":true,"status":"PASS","passing_atoms":["label:Code-Review\u003dMAX"],"failing_atoms":["label:Code-Review\u003dMIN"],"atom_explanations":{}}},{"name":"Workflow","description":"Approved for gate by core reviewer","status":"SATISFIED","is_legacy":false,"submittability_expression_result":{"expression":"label:Workflow\u003dMAX AND -label:Workflow\u003dMIN","fulfilled":true,"status":"PASS","passing_atoms":["label:Workflow\u003dMAX"],"failing_atoms":["label:Workflow\u003dMIN"],"atom_explanations":{}}}]}
