)]}'
{"id":"openstack%2Fopenstack-ansible-lxc_hosts~894558","triplet_id":"openstack%2Fopenstack-ansible-lxc_hosts~master~I1b08a45ccced87ecd200f3e7294165e922df39ff","project":"openstack/openstack-ansible-lxc_hosts","branch":"master","topic":"osa-bookworm","attention_set":{},"removed_from_attention_set":{"25023":{"account":{"_account_id":25023,"name":"Jonathan Rosser","email":"jonathan.rosser@rd.bbc.co.uk","username":"jrosser"},"last_update":"2023-10-10 16:09:33.000000000","reason":"Change was submitted"}},"hashtags":[],"change_id":"I1b08a45ccced87ecd200f3e7294165e922df39ff","subject":"Switch to native systemd-resolved from resolv.conf","status":"MERGED","created":"2023-09-11 16:14:19.000000000","updated":"2023-10-10 16:10:36.000000000","submitted":"2023-10-10 16:09:33.000000000","submitter":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"total_comment_count":0,"unresolved_comment_count":0,"has_review_started":true,"submission_id":"894558-osa-bookworm","meta_rev_id":"fe32c5c2be16dd8d72b38570387291954c7926ad","_number":894558,"virtual_id_number":894558,"owner":{"_account_id":25023,"name":"Jonathan Rosser","email":"jonathan.rosser@rd.bbc.co.uk","username":"jrosser"},"actions":{},"labels":{"Verified":{"approved":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"all":[{"value":0,"_account_id":31542,"name":"Andrew Bonney","email":"andrew.bonney@bbc.co.uk","username":"andrewbonney"},{"value":0,"_account_id":28619,"name":"Dmitriy Rabotyagov","email":"noonedeadpunk@gmail.com","username":"noonedeadpunk"},{"tag":"autogenerated:zuul:gate","value":2,"date":"2023-10-10 16:09:33.000000000","permitted_voting_range":{"min":2,"max":2},"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]}],"values":{"-2":"Fails","-1":"Doesn\u0027t seem to work"," 0":"No score","+1":"Works for me","+2":"Verified"},"description":"","default_value":0,"optional":true},"Code-Review":{"approved":{"_account_id":31542,"name":"Andrew Bonney","email":"andrew.bonney@bbc.co.uk","username":"andrewbonney"},"all":[{"value":2,"date":"2023-10-02 14:28:53.000000000","permitted_voting_range":{"min":2,"max":2},"_account_id":31542,"name":"Andrew Bonney","email":"andrew.bonney@bbc.co.uk","username":"andrewbonney"},{"value":2,"date":"2023-09-29 12:44:13.000000000","permitted_voting_range":{"min":2,"max":2},"_account_id":28619,"name":"Dmitriy Rabotyagov","email":"noonedeadpunk@gmail.com","username":"noonedeadpunk"},{"value":0,"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]}],"values":{"-2":"Do not merge","-1":"This patch needs further work before it can be merged"," 0":"No score","+1":"Looks good to me, but someone else must approve","+2":"Looks good to me (core reviewer)"},"description":"","default_value":0,"optional":true},"Workflow":{"approved":{"_account_id":31542,"name":"Andrew Bonney","email":"andrew.bonney@bbc.co.uk","username":"andrewbonney"},"all":[{"value":1,"date":"2023-10-10 13:34:59.000000000","permitted_voting_range":{"min":1,"max":1},"_account_id":31542,"name":"Andrew Bonney","email":"andrew.bonney@bbc.co.uk","username":"andrewbonney"},{"value":0,"_account_id":28619,"name":"Dmitriy Rabotyagov","email":"noonedeadpunk@gmail.com","username":"noonedeadpunk"},{"value":0,"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]}],"values":{"-1":"Work in progress"," 0":"Ready for reviews","+1":"Approved"},"description":"","default_value":0,"optional":true},"Backport-Candidate":{"all":[{"value":0,"_account_id":31542,"name":"Andrew Bonney","email":"andrew.bonney@bbc.co.uk","username":"andrewbonney"},{"value":0,"_account_id":28619,"name":"Dmitriy Rabotyagov","email":"noonedeadpunk@gmail.com","username":"noonedeadpunk"},{"value":0,"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]}],"values":{"-1":"Not A Backport Candidate"," 0":"Backport Review Needed","+1":"Proposed Backport"},"description":"","default_value":0,"optional":true}},"removable_reviewers":[],"reviewers":{"REVIEWER":[{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},{"_account_id":28619,"name":"Dmitriy Rabotyagov","email":"noonedeadpunk@gmail.com","username":"noonedeadpunk"},{"_account_id":31542,"name":"Andrew Bonney","email":"andrew.bonney@bbc.co.uk","username":"andrewbonney"}]},"pending_reviewers":{},"reviewer_updates":[{"updated":"2023-09-11 18:09:09.000000000","updated_by":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"reviewer":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"state":"REVIEWER"},{"updated":"2023-09-29 08:12:18.000000000","updated_by":{"_account_id":28619,"name":"Dmitriy Rabotyagov","email":"noonedeadpunk@gmail.com","username":"noonedeadpunk"},"reviewer":{"_account_id":28619,"name":"Dmitriy Rabotyagov","email":"noonedeadpunk@gmail.com","username":"noonedeadpunk"},"state":"REVIEWER"},{"updated":"2023-10-02 14:28:53.000000000","updated_by":{"_account_id":31542,"name":"Andrew Bonney","email":"andrew.bonney@bbc.co.uk","username":"andrewbonney"},"reviewer":{"_account_id":31542,"name":"Andrew Bonney","email":"andrew.bonney@bbc.co.uk","username":"andrewbonney"},"state":"REVIEWER"}],"messages":[{"id":"a19e3a3e128b11c194190d4a7ce75686c9ee4296","tag":"autogenerated:gerrit:newPatchSet","author":{"_account_id":25023,"name":"Jonathan Rosser","email":"jonathan.rosser@rd.bbc.co.uk","username":"jrosser"},"date":"2023-09-11 16:14:19.000000000","message":"Uploaded patch set 1.","accounts_in_message":[],"_revision_number":1},{"id":"6a3cb6dc11484d526c067add64e14b07e3a43713","tag":"autogenerated:zuul:check","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2023-09-11 18:09:09.000000000","message":"Patch Set 1: Verified-1\n\nBuild failed (check pipeline).  For information on how to proceed, see\nhttps://docs.opendev.org/opendev/infra-manual/latest/developers.html#automated-testing\n\nhttps://zuul.opendev.org/t/openstack/buildset/265ac59c2be7413dbc760868e4b94958\n\n- openstack-ansible-linters-ubuntu-jammy https://zuul.opendev.org/t/openstack/build/3095eef0ef5d4e4e95fef9340cd45e11 : SUCCESS in 13m 29s\n- openstack-tox-docs https://zuul.opendev.org/t/openstack/build/883551a8381d45b389451c7fa652beff : SUCCESS in 5m 12s\n- openstack-ansible-deploy-infra_lxc-centos-9-stream https://zuul.opendev.org/t/openstack/build/303f96628837457e985ba0e76f65d9b3 : SUCCESS in 1h 34m 35s (non-voting)\n- openstack-ansible-deploy-infra_lxc-rockylinux-9 https://zuul.opendev.org/t/openstack/build/dffd13d0ff464bad9d971d27f3a6ae56 : SUCCESS in 1h 28m 13s\n- openstack-ansible-deploy-infra_lxc_tls-rockylinux-9 https://zuul.opendev.org/t/openstack/build/2b03da20617941e092d2fccd6a59d312 : SUCCESS in 1h 41m 21s\n- openstack-ansible-deploy-infra_lxc-debian-bullseye https://zuul.opendev.org/t/openstack/build/ccc4a0078b7f4e1fb989ddeeba703c9e : FAILURE in 15m 49s\n- openstack-ansible-deploy-infra_lxc-ubuntu-jammy https://zuul.opendev.org/t/openstack/build/0453c42882934a4c8d3e1bf02040120d : FAILURE in 16m 33s\n- openstack-ansible-deploy-infra_lxc_stepca-ubuntu-jammy https://zuul.opendev.org/t/openstack/build/30b27fa6d83f42519bcf70fae52ab9f9 : FAILURE in 17m 50s\n- openstack-ansible-deploy-infra_lxc_tls-ubuntu-jammy https://zuul.opendev.org/t/openstack/build/b8dc8aba73534f54848e99c2d6843bbf : FAILURE in 21m 14s\n- openstack-ansible-upgrade-infra_lxc-rockylinux-9 https://zuul.opendev.org/t/openstack/build/c22667a36c3b40029bc2e2278e31aa74 : SUCCESS in 1h 46m 48s\n- openstack-ansible-upgrade-infra_lxc-ubuntu-jammy https://zuul.opendev.org/t/openstack/build/1c11609fc82f43c5b2300c692f03a808 : SUCCESS in 1h 46m 37s\n- openstack-ansible-deploy-hosts_lxc_zfs-ubuntu-jammy https://zuul.opendev.org/t/openstack/build/e517b5c778a14cb799f3d32dcdfc24af : FAILURE in 22m 50s\n- openstack-ansible-deploy-hosts_lxc_btrfs-ubuntu-jammy https://zuul.opendev.org/t/openstack/build/44b1d8ad5a5f4a23a7e6648799f31c74 : FAILURE in 20m 05s\n- openstack-ansible-deploy-hosts_lxc_overlayfs-ubuntu-jammy https://zuul.opendev.org/t/openstack/build/d8ab8a865c334478a6a61144a397d12d : FAILURE in 25m 52s\n- openstack-ansible-deploy-hosts_lxc_btrfs-debian-bullseye https://zuul.opendev.org/t/openstack/build/856a775820bc4139bb4c1f1651fa809a : FAILURE in 18m 11s\n- openstack-ansible-deploy-hosts_lxc_overlayfs-debian-bullseye https://zuul.opendev.org/t/openstack/build/e63033c0f2194af2a69a2501ca9dfdf9 : FAILURE in 18m 41s\n- openstack-ansible-deploy-hosts_lxc_overlayfs-rockylinux-9 https://zuul.opendev.org/t/openstack/build/765eb34636544f7b84f8d9d759930462 : SUCCESS in 28m 37s","accounts_in_message":[],"_revision_number":1},{"id":"2b41099013f719bb83689062f2f109da40ff9e14","tag":"autogenerated:gerrit:newPatchSet","author":{"_account_id":25023,"name":"Jonathan Rosser","email":"jonathan.rosser@rd.bbc.co.uk","username":"jrosser"},"date":"2023-09-11 18:30:02.000000000","message":"Uploaded patch set 2: Patch Set 1 was rebased.\n\nOutdated Votes:\n* Verified-1\n","accounts_in_message":[],"_revision_number":2},{"id":"6c13cf75ffa6aa90595e6190af9e84eb8e239dc2","tag":"autogenerated:zuul:check","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2023-09-11 20:36:24.000000000","message":"Patch Set 2: Verified+1\n\nBuild succeeded (check pipeline).\nhttps://zuul.opendev.org/t/openstack/buildset/3ef3eeac34614a0b8f800e54990203b0\n\n- openstack-ansible-linters-ubuntu-jammy https://zuul.opendev.org/t/openstack/build/4541b868041648878ff545c4d69b8cc0 : SUCCESS in 13m 35s\n- openstack-tox-docs https://zuul.opendev.org/t/openstack/build/3ca4df38a90945f6a9921e7216ce7010 : SUCCESS in 4m 21s\n- openstack-ansible-deploy-infra_lxc-centos-9-stream https://zuul.opendev.org/t/openstack/build/5cb19f36b07a42b0bf0bb2ccbc7b76af : SUCCESS in 1h 18m 51s (non-voting)\n- openstack-ansible-deploy-infra_lxc-rockylinux-9 https://zuul.opendev.org/t/openstack/build/9d7beab64cfa4158bc51d3f387934e00 : SUCCESS in 1h 30m 06s\n- openstack-ansible-deploy-infra_lxc_tls-rockylinux-9 https://zuul.opendev.org/t/openstack/build/9cf2fc746b1240718581c35665d86710 : SUCCESS in 1h 32m 25s\n- openstack-ansible-deploy-infra_lxc-debian-bullseye https://zuul.opendev.org/t/openstack/build/35ad7be453e6433088263e188e403854 : SUCCESS in 1h 14m 49s\n- openstack-ansible-deploy-infra_lxc-ubuntu-jammy https://zuul.opendev.org/t/openstack/build/2c501dd657354d7897e28e0ea40d8b05 : SUCCESS in 1h 13m 17s\n- openstack-ansible-deploy-infra_lxc_stepca-ubuntu-jammy https://zuul.opendev.org/t/openstack/build/2895862508864980a6ab0413014a6a5a : SUCCESS in 1h 21m 57s\n- openstack-ansible-deploy-infra_lxc_tls-ubuntu-jammy https://zuul.opendev.org/t/openstack/build/861583e53c5846dbb720a0fc1a8ed147 : SUCCESS in 1h 10m 02s\n- openstack-ansible-upgrade-infra_lxc-rockylinux-9 https://zuul.opendev.org/t/openstack/build/0e94a085940846048138520a82ce08ce : SUCCESS in 2h 02m 31s\n- openstack-ansible-upgrade-infra_lxc-ubuntu-jammy https://zuul.opendev.org/t/openstack/build/3bda3678fb9f42d8bc19f460a2de78b6 : SUCCESS in 1h 49m 25s\n- openstack-ansible-deploy-hosts_lxc_zfs-ubuntu-jammy https://zuul.opendev.org/t/openstack/build/80e75af763ff47b9b8a2c1fdf9269803 : SUCCESS in 28m 57s\n- openstack-ansible-deploy-hosts_lxc_btrfs-ubuntu-jammy https://zuul.opendev.org/t/openstack/build/6fe2a3b4b0cd4b1cb8cea2cc0eb57ab7 : SUCCESS in 27m 49s\n- openstack-ansible-deploy-hosts_lxc_overlayfs-ubuntu-jammy https://zuul.opendev.org/t/openstack/build/ef96bd3c7ecf41129415b3a46b24ebf7 : SUCCESS in 35m 33s\n- openstack-ansible-deploy-hosts_lxc_btrfs-debian-bullseye https://zuul.opendev.org/t/openstack/build/77f188192f2545a399bcd96da455296b : SUCCESS in 25m 17s\n- openstack-ansible-deploy-hosts_lxc_overlayfs-debian-bullseye https://zuul.opendev.org/t/openstack/build/679b2ac660394d8587bdf6370adf2b14 : SUCCESS in 32m 12s\n- openstack-ansible-deploy-hosts_lxc_overlayfs-rockylinux-9 https://zuul.opendev.org/t/openstack/build/a35b36f805de4d66ba84229dc7fbc30f : SUCCESS in 29m 24s","accounts_in_message":[],"_revision_number":2},{"id":"d24b94dd7d6ca0bddc3f7c38322bb0a37fa0f559","tag":"autogenerated:gerrit:newPatchSet","author":{"_account_id":25023,"name":"Jonathan Rosser","email":"jonathan.rosser@rd.bbc.co.uk","username":"jrosser"},"date":"2023-09-27 16:29:42.000000000","message":"Patch Set 3: Patch Set 2 was rebased\n\nOutdated Votes:\n* Verified+1\n","accounts_in_message":[],"_revision_number":3},{"id":"da4c9708356139430e41f061a4d599b4045c0bc8","tag":"autogenerated:zuul:check","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2023-09-27 18:18:01.000000000","message":"Patch Set 3: Verified-1\n\nBuild failed (check pipeline).  For information on how to proceed, see\nhttps://docs.opendev.org/opendev/infra-manual/latest/developers.html#automated-testing\n\nhttps://zuul.opendev.org/t/openstack/buildset/56a0b27bee614638a65f786934eebfad\n\n- openstack-ansible-linters-ubuntu-jammy https://zuul.opendev.org/t/openstack/build/7c0751fe6a2a492288850a35f9b016e4 : FAILURE in 12m 48s\n- openstack-tox-docs https://zuul.opendev.org/t/openstack/build/8431fb7377074397b0631c02513e79c8 : SUCCESS in 4m 33s\n- openstack-ansible-deploy-infra_lxc-centos-9-stream https://zuul.opendev.org/t/openstack/build/a39805f7f00941878d58bce015a8d4f5 : SUCCESS in 1h 20m 34s (non-voting)\n- openstack-ansible-deploy-infra_lxc-rockylinux-9 https://zuul.opendev.org/t/openstack/build/bce6fde37121486888c11c662f2b277a : FAILURE in 29m 25s\n- openstack-ansible-deploy-infra_lxc_tls-rockylinux-9 https://zuul.opendev.org/t/openstack/build/996604f3a0874d5699aa7323871242a5 : SUCCESS in 1h 37m 03s\n- openstack-ansible-deploy-infra_lxc-debian-bullseye https://zuul.opendev.org/t/openstack/build/41f3f3189ccd43c6903f2e32930d0337 : SUCCESS in 1h 26m 40s\n- openstack-ansible-deploy-infra_lxc-ubuntu-jammy https://zuul.opendev.org/t/openstack/build/ebf471db753f484abbce331e99df8018 : SUCCESS in 1h 16m 27s\n- openstack-ansible-deploy-infra_lxc_stepca-ubuntu-jammy https://zuul.opendev.org/t/openstack/build/c6ef24edba954f8d85875ae9c13999cc : SUCCESS in 1h 21m 28s\n- openstack-ansible-deploy-infra_lxc_tls-ubuntu-jammy https://zuul.opendev.org/t/openstack/build/29f42dbb9cb1433888e20924270af899 : SUCCESS in 1h 13m 06s\n- openstack-ansible-upgrade-infra_lxc-rockylinux-9 https://zuul.opendev.org/t/openstack/build/d75557e873494ca5a8f6d205d3f338fa : FAILURE in 19m 49s\n- openstack-ansible-upgrade-infra_lxc-ubuntu-jammy https://zuul.opendev.org/t/openstack/build/178cf5bf53754eeb993323ab9607855a : SUCCESS in 1h 41m 47s\n- openstack-ansible-deploy-hosts_lxc_zfs-ubuntu-jammy https://zuul.opendev.org/t/openstack/build/0c83756281914b67b024ed0addc72243 : SUCCESS in 30m 09s\n- openstack-ansible-deploy-hosts_lxc_btrfs-ubuntu-jammy https://zuul.opendev.org/t/openstack/build/3713aeaebc96436dbdfc69830b1a778f : SUCCESS in 29m 00s\n- openstack-ansible-deploy-hosts_lxc_overlayfs-ubuntu-jammy https://zuul.opendev.org/t/openstack/build/c825ca78341b4f0d8cd70015e47539c2 : SUCCESS in 26m 27s\n- openstack-ansible-deploy-hosts_lxc_btrfs-debian-bullseye https://zuul.opendev.org/t/openstack/build/efcf2f546d4e482cbd91c429cbc65955 : SUCCESS in 32m 05s\n- openstack-ansible-deploy-hosts_lxc_overlayfs-debian-bullseye https://zuul.opendev.org/t/openstack/build/96fae9953cc249419db49e4d7285b27d : SUCCESS in 32m 01s\n- openstack-ansible-deploy-hosts_lxc_overlayfs-rockylinux-9 https://zuul.opendev.org/t/openstack/build/93f2dc08a34c421bb6557c833f870711 : SUCCESS in 36m 51s","accounts_in_message":[],"_revision_number":3},{"id":"c371c65280aa4b03baa056e880b74c7915056522","tag":"autogenerated:gerrit:newPatchSet","author":{"_account_id":25023,"name":"Jonathan Rosser","email":"jonathan.rosser@rd.bbc.co.uk","username":"jrosser"},"date":"2023-09-28 09:27:01.000000000","message":"Uploaded patch set 4: Patch Set 3 was rebased.\n\nOutdated Votes:\n* Verified-1\n","accounts_in_message":[],"_revision_number":4},{"id":"df87c2e0a1f37f8130a7839d66299db355f94aed","tag":"autogenerated:zuul:check","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2023-09-28 12:09:55.000000000","message":"Patch Set 4: Verified+1\n\nBuild succeeded (check pipeline).\nhttps://zuul.opendev.org/t/openstack/buildset/8ed26f6ba2a445dfb2d29a322e0c1b0a\n\n- openstack-ansible-linters-ubuntu-jammy https://zuul.opendev.org/t/openstack/build/dbec305b951e4d3a9bc87d1f6ea57d56 : SUCCESS in 18m 33s\n- openstack-tox-docs https://zuul.opendev.org/t/openstack/build/0d2b0671e3c74b06b3ecef00a2c7ba8b : SUCCESS in 4m 50s\n- openstack-ansible-deploy-infra_lxc-centos-9-stream https://zuul.opendev.org/t/openstack/build/f491f4f33861472bbc0d2946ce961245 : SUCCESS in 1h 50m 14s (non-voting)\n- openstack-ansible-deploy-infra_lxc-rockylinux-9 https://zuul.opendev.org/t/openstack/build/183db92176f94a038378b99f56b951f0 : SUCCESS in 1h 38m 36s\n- openstack-ansible-deploy-infra_lxc_tls-rockylinux-9 https://zuul.opendev.org/t/openstack/build/80cd0032b152427698914fa21cbb537e : SUCCESS in 1h 39m 46s\n- openstack-ansible-deploy-infra_lxc-debian-bullseye https://zuul.opendev.org/t/openstack/build/5058939d4dab46369b008ff0997912b5 : SUCCESS in 1h 28m 26s\n- openstack-ansible-deploy-infra_lxc-ubuntu-jammy https://zuul.opendev.org/t/openstack/build/3d11219b1e7242ac87ee3ad74a2d1ca1 : SUCCESS in 1h 18m 52s\n- openstack-ansible-deploy-infra_lxc_stepca-ubuntu-jammy https://zuul.opendev.org/t/openstack/build/19cfa0ac5bc343bf9382cb47abc89dc4 : SUCCESS in 1h 14m 32s\n- openstack-ansible-deploy-infra_lxc_tls-ubuntu-jammy https://zuul.opendev.org/t/openstack/build/dc0aacb3216e4799b7b7a88171a83333 : SUCCESS in 1h 26m 39s\n- openstack-ansible-upgrade-infra_lxc-rockylinux-9 https://zuul.opendev.org/t/openstack/build/af363d72f2694053ade9e6ac4073febd : SUCCESS in 2h 34m 59s\n- openstack-ansible-upgrade-infra_lxc-ubuntu-jammy https://zuul.opendev.org/t/openstack/build/3acc332886ac4b08aec1385f4d3b2e3b : SUCCESS in 1h 40m 42s\n- openstack-ansible-deploy-hosts_lxc_zfs-ubuntu-jammy https://zuul.opendev.org/t/openstack/build/076cb3885b0646b2a0f1441606496a5b : SUCCESS in 33m 57s\n- openstack-ansible-deploy-hosts_lxc_btrfs-ubuntu-jammy https://zuul.opendev.org/t/openstack/build/d07e8eba9111434981612d19b9f9d9ae : SUCCESS in 32m 02s\n- openstack-ansible-deploy-hosts_lxc_overlayfs-ubuntu-jammy https://zuul.opendev.org/t/openstack/build/4ca31093ac9f47a6bf2a882d262e452d : SUCCESS in 31m 22s\n- openstack-ansible-deploy-hosts_lxc_btrfs-debian-bullseye https://zuul.opendev.org/t/openstack/build/3c10f778a55642e9b95b22260ef4a897 : SUCCESS in 32m 07s\n- openstack-ansible-deploy-hosts_lxc_overlayfs-debian-bullseye https://zuul.opendev.org/t/openstack/build/63f664f2b54c4a69949cd5af8954465b : SUCCESS in 34m 01s\n- openstack-ansible-deploy-hosts_lxc_overlayfs-rockylinux-9 https://zuul.opendev.org/t/openstack/build/4fa8c4bdf38b4308b84937708d3fcd8c : SUCCESS in 36m 24s","accounts_in_message":[],"_revision_number":4},{"id":"cb2f10630d2a86837652384604577f46d8fee573","author":{"_account_id":28619,"name":"Dmitriy Rabotyagov","email":"noonedeadpunk@gmail.com","username":"noonedeadpunk"},"date":"2023-09-29 08:12:18.000000000","message":"Patch Set 4: Code-Review+2","accounts_in_message":[],"_revision_number":4},{"id":"f4c6354bb4b6ded7977f1c7457187c8862a8941c","tag":"autogenerated:gerrit:newPatchSet","author":{"_account_id":25023,"name":"Jonathan Rosser","email":"jonathan.rosser@rd.bbc.co.uk","username":"jrosser"},"date":"2023-09-29 12:44:13.000000000","message":"Patch Set 5: Patch Set 4 was rebased\n\nCopied Votes:\n* Code-Review+2 (copy condition: \"changekind:TRIVIAL_REBASE OR is:MIN\")\n\nOutdated Votes:\n* Verified+1\n","accounts_in_message":[],"_revision_number":5},{"id":"4ceec9b158554a5b965e9b301e5c668a1468f595","tag":"autogenerated:zuul:check","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2023-09-29 14:46:54.000000000","message":"Patch Set 5: Verified+1\n\nBuild succeeded (check pipeline).\nhttps://zuul.opendev.org/t/openstack/buildset/0b1b3cc0a95a4173ba1cd5c53208340f\n\n- openstack-ansible-linters-ubuntu-jammy https://zuul.opendev.org/t/openstack/build/0516115283ad4e2c98f84e5141621d4a : SUCCESS in 16m 28s\n- openstack-tox-docs https://zuul.opendev.org/t/openstack/build/0325cf70c76e474fb33d9a5b0bfc3055 : SUCCESS in 4m 56s\n- openstack-ansible-deploy-infra_lxc-centos-9-stream https://zuul.opendev.org/t/openstack/build/8e477fe7526945eebbbefb6b1ac8b6d9 : SUCCESS in 1h 31m 44s (non-voting)\n- openstack-ansible-deploy-infra_lxc-rockylinux-9 https://zuul.opendev.org/t/openstack/build/7ebf81308a8c455d9d66e7461c8265b6 : SUCCESS in 1h 43m 01s\n- openstack-ansible-deploy-infra_lxc_tls-rockylinux-9 https://zuul.opendev.org/t/openstack/build/453ff40314474a8fad6a6aa6004a72af : SUCCESS in 1h 38m 20s\n- openstack-ansible-deploy-infra_lxc-debian-bullseye https://zuul.opendev.org/t/openstack/build/5950fccfacad443b8e80dc99891d220c : SUCCESS in 1h 07m 11s\n- openstack-ansible-deploy-infra_lxc-ubuntu-jammy https://zuul.opendev.org/t/openstack/build/e6071c6a03924f3c9e949b52e736ff12 : SUCCESS in 1h 23m 50s\n- openstack-ansible-deploy-infra_lxc_stepca-ubuntu-jammy https://zuul.opendev.org/t/openstack/build/ad3a23d586b34b1288648ba1083f4f38 : SUCCESS in 1h 20m 09s\n- openstack-ansible-deploy-infra_lxc_tls-ubuntu-jammy https://zuul.opendev.org/t/openstack/build/eea53322b7934c9193671a30a8ac6e17 : SUCCESS in 1h 25m 56s\n- openstack-ansible-upgrade-infra_lxc-rockylinux-9 https://zuul.opendev.org/t/openstack/build/4d42e18558fc49dbbeae811c3369c502 : SUCCESS in 1h 50m 59s\n- openstack-ansible-upgrade-infra_lxc-ubuntu-jammy https://zuul.opendev.org/t/openstack/build/fe997df67f6047b9afdde2720b284e22 : SUCCESS in 1h 49m 51s\n- openstack-ansible-deploy-hosts_lxc_zfs-ubuntu-jammy https://zuul.opendev.org/t/openstack/build/dceb8f489b80405abe432ee08775e05d : SUCCESS in 30m 04s\n- openstack-ansible-deploy-hosts_lxc_btrfs-ubuntu-jammy https://zuul.opendev.org/t/openstack/build/dbcc8f0af1ca45a6be0272e45d6d80be : SUCCESS in 27m 47s\n- openstack-ansible-deploy-hosts_lxc_overlayfs-ubuntu-jammy https://zuul.opendev.org/t/openstack/build/f4709b973214412b9c6df5f2e5e94722 : SUCCESS in 28m 32s\n- openstack-ansible-deploy-hosts_lxc_btrfs-debian-bullseye https://zuul.opendev.org/t/openstack/build/02f819e848664afead1369a0158b2dec : SUCCESS in 32m 24s\n- openstack-ansible-deploy-hosts_lxc_overlayfs-debian-bullseye https://zuul.opendev.org/t/openstack/build/4f03005d036a420fa0f1c067647b34e3 : SUCCESS in 27m 30s\n- openstack-ansible-deploy-hosts_lxc_overlayfs-rockylinux-9 https://zuul.opendev.org/t/openstack/build/3d8c6706c347469eb32e18c902049bf5 : SUCCESS in 30m 29s","accounts_in_message":[],"_revision_number":5},{"id":"9566347c383be7c89f6d436795b41230a32f7e3a","author":{"_account_id":31542,"name":"Andrew Bonney","email":"andrew.bonney@bbc.co.uk","username":"andrewbonney"},"date":"2023-10-02 14:28:53.000000000","message":"Patch Set 5: Code-Review+2","accounts_in_message":[],"_revision_number":5},{"id":"d88d5b286af2aabe919d54ed05a4bc87e6075006","author":{"_account_id":31542,"name":"Andrew Bonney","email":"andrew.bonney@bbc.co.uk","username":"andrewbonney"},"date":"2023-10-10 13:34:59.000000000","message":"Patch Set 5: Workflow+1","accounts_in_message":[],"_revision_number":5},{"id":"817c60c3c2c4ddc9d93b1a4e41b23e93f980b56d","tag":"autogenerated:zuul:gate","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2023-10-10 13:37:11.000000000","message":"Patch Set 5: -Verified\n\nStarting gate jobs.","accounts_in_message":[],"_revision_number":5},{"id":"ce27241291eaf2575856f7cd541d9618d9c9851e","tag":"autogenerated:zuul:gate","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2023-10-10 16:09:33.000000000","message":"Patch Set 5: Verified+2\n\nBuild succeeded (gate pipeline).\nhttps://zuul.opendev.org/t/openstack/buildset/b3cbdc4d1568476d9c94eba4c49a9f54\n\n- openstack-ansible-linters-ubuntu-jammy https://zuul.opendev.org/t/openstack/build/1de8a43f702e42bca345c064f941a66d : SUCCESS in 12m 01s\n- openstack-tox-docs https://zuul.opendev.org/t/openstack/build/a39902d586494c089e8828a77b83b85f : SUCCESS in 4m 33s\n- openstack-ansible-deploy-infra_lxc-rockylinux-9 https://zuul.opendev.org/t/openstack/build/71593c6a52d24e97a28e70a8c67369f8 : SUCCESS in 1h 33m 35s\n- openstack-ansible-deploy-infra_lxc_tls-rockylinux-9 https://zuul.opendev.org/t/openstack/build/0bd4094a5c1e44fdbb564d8d91788ae6 : SUCCESS in 1h 48m 22s\n- openstack-ansible-deploy-infra_lxc-debian-bullseye https://zuul.opendev.org/t/openstack/build/40b57069cb0a4f719f3c2cf7fd54eb17 : SUCCESS in 1h 04m 58s\n- openstack-ansible-deploy-infra_lxc-ubuntu-jammy https://zuul.opendev.org/t/openstack/build/ad8c715ca08a40b7a08ddaf3823a7db8 : SUCCESS in 1h 06m 33s\n- openstack-ansible-deploy-infra_lxc_stepca-ubuntu-jammy https://zuul.opendev.org/t/openstack/build/f12b64c878c147f484ba94c925c24e12 : SUCCESS in 1h 34m 17s\n- openstack-ansible-deploy-infra_lxc_tls-ubuntu-jammy https://zuul.opendev.org/t/openstack/build/f1f5171b2068440480f78f1d63d8c2b2 : SUCCESS in 1h 14m 07s\n- openstack-ansible-upgrade-infra_lxc-rockylinux-9 https://zuul.opendev.org/t/openstack/build/94af9dad44d3444aa9e47d6aeeebacf6 : SUCCESS in 1h 58m 33s\n- openstack-ansible-upgrade-infra_lxc-ubuntu-jammy https://zuul.opendev.org/t/openstack/build/d0251576e5d7430097671dfde387c5a7 : SUCCESS in 1h 21m 39s\n- openstack-ansible-deploy-hosts_lxc_zfs-ubuntu-jammy https://zuul.opendev.org/t/openstack/build/19bab585c6c146bb8730614e14669d90 : SUCCESS in 31m 55s\n- openstack-ansible-deploy-hosts_lxc_btrfs-ubuntu-jammy https://zuul.opendev.org/t/openstack/build/684e092042274ca4b216b04021b98f65 : SUCCESS in 33m 08s\n- openstack-ansible-deploy-hosts_lxc_overlayfs-ubuntu-jammy https://zuul.opendev.org/t/openstack/build/323a3d1f581140d6ad1a293ec3cd0078 : SUCCESS in 28m 36s\n- openstack-ansible-deploy-hosts_lxc_btrfs-debian-bullseye https://zuul.opendev.org/t/openstack/build/04c22e1e6fce41ccbe74aff38839a65c : SUCCESS in 27m 27s\n- openstack-ansible-deploy-hosts_lxc_overlayfs-debian-bullseye https://zuul.opendev.org/t/openstack/build/55c670d0e7744bbb99c5684b18ac2cd1 : SUCCESS in 27m 25s\n- openstack-ansible-deploy-hosts_lxc_overlayfs-rockylinux-9 https://zuul.opendev.org/t/openstack/build/b2911c2fc2c24c4c910b230799d6dded : SUCCESS in 29m 14s","accounts_in_message":[],"_revision_number":5},{"id":"822be0bb583fa02c2ac3e4c9bea291532384c51e","tag":"autogenerated:gerrit:merged","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2023-10-10 16:09:33.000000000","message":"Change has been successfully merged","accounts_in_message":[],"_revision_number":5},{"id":"fe32c5c2be16dd8d72b38570387291954c7926ad","tag":"autogenerated:zuul:promote","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2023-10-10 16:10:36.000000000","message":"Patch Set 5:\n\nBuild succeeded (promote pipeline).\nhttps://zuul.opendev.org/t/openstack/buildset/268bf6c6cd73403482a04d0ebf39ceae\n\n- promote-openstack-tox-docs https://zuul.opendev.org/t/openstack/build/62c06c261c6948f0a8d07e9669662d13 : SUCCESS in 43s","accounts_in_message":[],"_revision_number":5}],"current_revision_number":5,"current_revision":"39542e2c6594d3181108e383e59a423186d16c39","revisions":{"4248b82ce053d308c1b1c0bd09283c6145da99c4":{"kind":"REWORK","_number":1,"created":"2023-09-11 16:14:19.000000000","uploader":{"_account_id":25023,"name":"Jonathan Rosser","email":"jonathan.rosser@rd.bbc.co.uk","username":"jrosser"},"ref":"refs/changes/58/894558/1","fetch":{"anonymous http":{"url":"https://review.opendev.org/openstack/openstack-ansible-lxc_hosts","ref":"refs/changes/58/894558/1","commands":{"Checkout":"git fetch https://review.opendev.org/openstack/openstack-ansible-lxc_hosts refs/changes/58/894558/1 \u0026\u0026 git checkout FETCH_HEAD","Cherry Pick":"git fetch https://review.opendev.org/openstack/openstack-ansible-lxc_hosts refs/changes/58/894558/1 \u0026\u0026 git cherry-pick FETCH_HEAD","Format Patch":"git fetch https://review.opendev.org/openstack/openstack-ansible-lxc_hosts refs/changes/58/894558/1 \u0026\u0026 git format-patch -1 --stdout FETCH_HEAD","Pull":"git pull https://review.opendev.org/openstack/openstack-ansible-lxc_hosts refs/changes/58/894558/1"}}},"commit":{"parents":[{"commit":"5d79bec6a175acfa37c960c15235d13fbe7147c6","subject":"Add ca-certificates into debian base image during debootstrap.","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/openstack-ansible-lxc_hosts/commit/5d79bec6a175acfa37c960c15235d13fbe7147c6"}]}],"author":{"name":"Jonathan Rosser","email":"jonathan.rosser@rd.bbc.co.uk","date":"2023-09-11 16:08:00.000000000","tz":60},"committer":{"name":"Jonathan Rosser","email":"jonathan.rosser@rd.bbc.co.uk","date":"2023-09-11 16:13:28.000000000","tz":60},"subject":"Switch to native systemd-resolved from resolv.conf","message":"Switch to native systemd-resolved from resolv.conf\n\nRather than edit resolv.conf, use the recommended method of\noperation for systemd-resolved and configure the dns server\nthrough a resolved drop-in.\n\nChange-Id: I1b08a45ccced87ecd200f3e7294165e922df39ff\n","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/openstack-ansible-lxc_hosts/commit/4248b82ce053d308c1b1c0bd09283c6145da99c4"}],"resolve_conflicts_web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/openstack-ansible-lxc_hosts/commit/4248b82ce053d308c1b1c0bd09283c6145da99c4"}]},"branch":"refs/heads/master"},"59b1b0e752ad8fffca9a118daa54252fcfb83a4b":{"kind":"TRIVIAL_REBASE","_number":2,"created":"2023-09-11 18:30:02.000000000","uploader":{"_account_id":25023,"name":"Jonathan Rosser","email":"jonathan.rosser@rd.bbc.co.uk","username":"jrosser"},"ref":"refs/changes/58/894558/2","fetch":{"anonymous http":{"url":"https://review.opendev.org/openstack/openstack-ansible-lxc_hosts","ref":"refs/changes/58/894558/2","commands":{"Checkout":"git fetch https://review.opendev.org/openstack/openstack-ansible-lxc_hosts refs/changes/58/894558/2 \u0026\u0026 git checkout FETCH_HEAD","Cherry Pick":"git fetch https://review.opendev.org/openstack/openstack-ansible-lxc_hosts refs/changes/58/894558/2 \u0026\u0026 git cherry-pick FETCH_HEAD","Format Patch":"git fetch https://review.opendev.org/openstack/openstack-ansible-lxc_hosts refs/changes/58/894558/2 \u0026\u0026 git format-patch -1 --stdout FETCH_HEAD","Pull":"git pull https://review.opendev.org/openstack/openstack-ansible-lxc_hosts refs/changes/58/894558/2"}}},"commit":{"parents":[{"commit":"cce5613c75113781e59082be401249214a37ccaf","subject":"Add ca-certificates into debian base image during debootstrap.","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/openstack-ansible-lxc_hosts/commit/cce5613c75113781e59082be401249214a37ccaf"}]}],"author":{"name":"Jonathan Rosser","email":"jonathan.rosser@rd.bbc.co.uk","date":"2023-09-11 16:08:00.000000000","tz":60},"committer":{"name":"Jonathan Rosser","email":"jonathan.rosser@rd.bbc.co.uk","date":"2023-09-11 18:29:34.000000000","tz":60},"subject":"Switch to native systemd-resolved from resolv.conf","message":"Switch to native systemd-resolved from resolv.conf\n\nRather than edit resolv.conf, use the recommended method of\noperation for systemd-resolved and configure the dns server\nthrough a resolved drop-in.\n\nChange-Id: I1b08a45ccced87ecd200f3e7294165e922df39ff\n","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/openstack-ansible-lxc_hosts/commit/59b1b0e752ad8fffca9a118daa54252fcfb83a4b"}],"resolve_conflicts_web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/openstack-ansible-lxc_hosts/commit/59b1b0e752ad8fffca9a118daa54252fcfb83a4b"}]},"branch":"refs/heads/master"},"37483ba125f45fac149f33287b807d7d3a54f0f2":{"kind":"TRIVIAL_REBASE","_number":3,"created":"2023-09-27 16:29:42.000000000","uploader":{"_account_id":25023,"name":"Jonathan Rosser","email":"jonathan.rosser@rd.bbc.co.uk","username":"jrosser"},"ref":"refs/changes/58/894558/3","fetch":{"anonymous http":{"url":"https://review.opendev.org/openstack/openstack-ansible-lxc_hosts","ref":"refs/changes/58/894558/3","commands":{"Checkout":"git fetch https://review.opendev.org/openstack/openstack-ansible-lxc_hosts refs/changes/58/894558/3 \u0026\u0026 git checkout FETCH_HEAD","Cherry Pick":"git fetch https://review.opendev.org/openstack/openstack-ansible-lxc_hosts refs/changes/58/894558/3 \u0026\u0026 git cherry-pick FETCH_HEAD","Format Patch":"git fetch https://review.opendev.org/openstack/openstack-ansible-lxc_hosts refs/changes/58/894558/3 \u0026\u0026 git format-patch -1 --stdout FETCH_HEAD","Pull":"git pull https://review.opendev.org/openstack/openstack-ansible-lxc_hosts refs/changes/58/894558/3"}}},"commit":{"parents":[{"commit":"c4e272074966c00052cfe74545000ecca9303b18","subject":"Add ca-certificates into debian base image during debootstrap.","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/openstack-ansible-lxc_hosts/commit/c4e272074966c00052cfe74545000ecca9303b18"}]}],"author":{"name":"Jonathan Rosser","email":"jonathan.rosser@rd.bbc.co.uk","date":"2023-09-11 16:08:00.000000000","tz":60},"committer":{"name":"Jonathan Rosser","email":"jonathan.rosser@rd.bbc.co.uk","date":"2023-09-27 16:29:42.000000000","tz":0},"subject":"Switch to native systemd-resolved from resolv.conf","message":"Switch to native systemd-resolved from resolv.conf\n\nRather than edit resolv.conf, use the recommended method of\noperation for systemd-resolved and configure the dns server\nthrough a resolved drop-in.\n\nChange-Id: I1b08a45ccced87ecd200f3e7294165e922df39ff\n","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/openstack-ansible-lxc_hosts/commit/37483ba125f45fac149f33287b807d7d3a54f0f2"}],"resolve_conflicts_web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/openstack-ansible-lxc_hosts/commit/37483ba125f45fac149f33287b807d7d3a54f0f2"}]},"branch":"refs/heads/master","description":"Rebase"},"34ca02a1d837163362f60dd2d681875ef2ec8ed3":{"kind":"TRIVIAL_REBASE","_number":4,"created":"2023-09-28 09:27:01.000000000","uploader":{"_account_id":25023,"name":"Jonathan Rosser","email":"jonathan.rosser@rd.bbc.co.uk","username":"jrosser"},"ref":"refs/changes/58/894558/4","fetch":{"anonymous http":{"url":"https://review.opendev.org/openstack/openstack-ansible-lxc_hosts","ref":"refs/changes/58/894558/4","commands":{"Checkout":"git fetch https://review.opendev.org/openstack/openstack-ansible-lxc_hosts refs/changes/58/894558/4 \u0026\u0026 git checkout FETCH_HEAD","Cherry Pick":"git fetch https://review.opendev.org/openstack/openstack-ansible-lxc_hosts refs/changes/58/894558/4 \u0026\u0026 git cherry-pick FETCH_HEAD","Format Patch":"git fetch https://review.opendev.org/openstack/openstack-ansible-lxc_hosts refs/changes/58/894558/4 \u0026\u0026 git format-patch -1 --stdout FETCH_HEAD","Pull":"git pull https://review.opendev.org/openstack/openstack-ansible-lxc_hosts refs/changes/58/894558/4"}}},"commit":{"parents":[{"commit":"efbbcedc1d26b435552e02e8e355fb9b0e53e916","subject":"Add ca-certificates into debian base image during debootstrap.","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/openstack-ansible-lxc_hosts/commit/efbbcedc1d26b435552e02e8e355fb9b0e53e916"}]}],"author":{"name":"Jonathan Rosser","email":"jonathan.rosser@rd.bbc.co.uk","date":"2023-09-11 16:08:00.000000000","tz":60},"committer":{"name":"Jonathan Rosser","email":"jonathan.rosser@rd.bbc.co.uk","date":"2023-09-28 09:26:37.000000000","tz":60},"subject":"Switch to native systemd-resolved from resolv.conf","message":"Switch to native systemd-resolved from resolv.conf\n\nRather than edit resolv.conf, use the recommended method of\noperation for systemd-resolved and configure the dns server\nthrough a resolved drop-in.\n\nChange-Id: I1b08a45ccced87ecd200f3e7294165e922df39ff\n","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/openstack-ansible-lxc_hosts/commit/34ca02a1d837163362f60dd2d681875ef2ec8ed3"}],"resolve_conflicts_web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/openstack-ansible-lxc_hosts/commit/34ca02a1d837163362f60dd2d681875ef2ec8ed3"}]},"branch":"refs/heads/master"},"39542e2c6594d3181108e383e59a423186d16c39":{"kind":"TRIVIAL_REBASE","_number":5,"created":"2023-09-29 12:44:13.000000000","uploader":{"_account_id":25023,"name":"Jonathan Rosser","email":"jonathan.rosser@rd.bbc.co.uk","username":"jrosser"},"ref":"refs/changes/58/894558/5","fetch":{"anonymous http":{"url":"https://review.opendev.org/openstack/openstack-ansible-lxc_hosts","ref":"refs/changes/58/894558/5","commands":{"Checkout":"git fetch https://review.opendev.org/openstack/openstack-ansible-lxc_hosts refs/changes/58/894558/5 \u0026\u0026 git checkout FETCH_HEAD","Cherry Pick":"git fetch https://review.opendev.org/openstack/openstack-ansible-lxc_hosts refs/changes/58/894558/5 \u0026\u0026 git cherry-pick FETCH_HEAD","Format Patch":"git fetch https://review.opendev.org/openstack/openstack-ansible-lxc_hosts refs/changes/58/894558/5 \u0026\u0026 git format-patch -1 --stdout FETCH_HEAD","Pull":"git pull https://review.opendev.org/openstack/openstack-ansible-lxc_hosts refs/changes/58/894558/5"}}},"commit":{"parents":[{"commit":"03dc91fa85d3c717ec0de00ba894999f62234193","subject":"Add ca-certificates into debian base image during debootstrap.","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/openstack-ansible-lxc_hosts/commit/03dc91fa85d3c717ec0de00ba894999f62234193"}]}],"author":{"name":"Jonathan Rosser","email":"jonathan.rosser@rd.bbc.co.uk","date":"2023-09-11 16:08:00.000000000","tz":60},"committer":{"name":"Jonathan Rosser","email":"jonathan.rosser@rd.bbc.co.uk","date":"2023-09-29 12:44:13.000000000","tz":0},"subject":"Switch to native systemd-resolved from resolv.conf","message":"Switch to native systemd-resolved from resolv.conf\n\nRather than edit resolv.conf, use the recommended method of\noperation for systemd-resolved and configure the dns server\nthrough a resolved drop-in.\n\nChange-Id: I1b08a45ccced87ecd200f3e7294165e922df39ff\n","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/openstack-ansible-lxc_hosts/commit/39542e2c6594d3181108e383e59a423186d16c39"}],"resolve_conflicts_web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/openstack-ansible-lxc_hosts/commit/39542e2c6594d3181108e383e59a423186d16c39"}]},"branch":"refs/heads/master","description":"Rebase"}},"requirements":[],"submit_records":[{"rule_name":"gerrit~DefaultSubmitRule","status":"CLOSED","labels":[{"label":"Verified","status":"MAY","applied_by":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]}},{"label":"Code-Review","status":"MAY","applied_by":{"_account_id":31542,"name":"Andrew Bonney","email":"andrew.bonney@bbc.co.uk","username":"andrewbonney"}},{"label":"Workflow","status":"MAY","applied_by":{"_account_id":31542,"name":"Andrew Bonney","email":"andrew.bonney@bbc.co.uk","username":"andrewbonney"}},{"label":"Backport-Candidate","status":"MAY"}]}],"submit_requirements":[{"name":"Verified","description":"Verified in gate by CI","status":"SATISFIED","is_legacy":false,"submittability_expression_result":{"expression":"label:Verified\u003dMAX AND -label:Verified\u003dMIN","fulfilled":true,"status":"PASS","passing_atoms":["label:Verified\u003dMAX"],"failing_atoms":["label:Verified\u003dMIN"],"atom_explanations":{}}},{"name":"Backport-Candidate","description":"Backport candidate status","status":"NOT_APPLICABLE","is_legacy":false,"applicability_expression_result":{"fulfilled":false,"status":"FAIL"},"submittability_expression_result":{"expression":"is:true","fulfilled":true,"status":"NOT_EVALUATED","passing_atoms":[],"failing_atoms":[],"atom_explanations":{}}},{"name":"Code-Review","description":"Code reviewed by core reviewer","status":"SATISFIED","is_legacy":false,"submittability_expression_result":{"expression":"label:Code-Review\u003dMAX AND -label:Code-Review\u003dMIN","fulfilled":true,"status":"PASS","passing_atoms":["label:Code-Review\u003dMAX"],"failing_atoms":["label:Code-Review\u003dMIN"],"atom_explanations":{}}},{"name":"Workflow","description":"Approved for gate by core reviewer","status":"SATISFIED","is_legacy":false,"submittability_expression_result":{"expression":"label:Workflow\u003dMAX AND -label:Workflow\u003dMIN","fulfilled":true,"status":"PASS","passing_atoms":["label:Workflow\u003dMAX"],"failing_atoms":["label:Workflow\u003dMIN"],"atom_explanations":{}}}]}
