)]}'
{"id":"openstack%2Fopenstack-ansible-lxc_hosts~896750","triplet_id":"openstack%2Fopenstack-ansible-lxc_hosts~master~I2466a9959bc93da754e11d8b9c6fb5d90f64163f","project":"openstack/openstack-ansible-lxc_hosts","branch":"master","attention_set":{},"removed_from_attention_set":{"25023":{"account":{"_account_id":25023,"name":"Jonathan Rosser","email":"jonathan.rosser@rd.bbc.co.uk","username":"jrosser"},"last_update":"2023-10-02 16:42:03.000000000","reason":"\u003cGERRIT_ACCOUNT_25023\u003e replied on the change","reason_account":{"_account_id":25023,"name":"Jonathan Rosser","email":"jonathan.rosser@rd.bbc.co.uk","username":"jrosser"}}},"hashtags":[],"change_id":"I2466a9959bc93da754e11d8b9c6fb5d90f64163f","subject":"Fix linter error","status":"MERGED","created":"2023-09-28 09:00:16.000000000","updated":"2023-10-02 20:53:36.000000000","submitted":"2023-10-02 20:52:38.000000000","submitter":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"total_comment_count":2,"unresolved_comment_count":0,"has_review_started":true,"submission_id":"896750","meta_rev_id":"42adfb96a4d091e8c396c0edd91439c35e54da08","_number":896750,"virtual_id_number":896750,"owner":{"_account_id":25023,"name":"Jonathan Rosser","email":"jonathan.rosser@rd.bbc.co.uk","username":"jrosser"},"actions":{},"labels":{"Verified":{"approved":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"all":[{"value":0,"_account_id":31542,"name":"Andrew Bonney","email":"andrew.bonney@bbc.co.uk","username":"andrewbonney"},{"value":0,"_account_id":28619,"name":"Dmitriy Rabotyagov","email":"noonedeadpunk@gmail.com","username":"noonedeadpunk"},{"tag":"autogenerated:zuul:gate","value":2,"date":"2023-10-02 20:52:38.000000000","permitted_voting_range":{"min":2,"max":2},"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]}],"values":{"-2":"Fails","-1":"Doesn\u0027t seem to work"," 0":"No score","+1":"Works for me","+2":"Verified"},"description":"","default_value":0,"optional":true},"Code-Review":{"approved":{"_account_id":31542,"name":"Andrew Bonney","email":"andrew.bonney@bbc.co.uk","username":"andrewbonney"},"all":[{"value":2,"date":"2023-10-02 14:30:32.000000000","permitted_voting_range":{"min":2,"max":2},"_account_id":31542,"name":"Andrew Bonney","email":"andrew.bonney@bbc.co.uk","username":"andrewbonney"},{"value":2,"date":"2023-09-28 09:09:04.000000000","permitted_voting_range":{"min":2,"max":2},"_account_id":28619,"name":"Dmitriy Rabotyagov","email":"noonedeadpunk@gmail.com","username":"noonedeadpunk"},{"value":0,"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]}],"values":{"-2":"Do not merge","-1":"This patch needs further work before it can be merged"," 0":"No score","+1":"Looks good to me, but someone else must approve","+2":"Looks good to me (core reviewer)"},"description":"","default_value":0,"optional":true},"Workflow":{"approved":{"_account_id":31542,"name":"Andrew Bonney","email":"andrew.bonney@bbc.co.uk","username":"andrewbonney"},"all":[{"value":1,"date":"2023-10-02 14:30:32.000000000","permitted_voting_range":{"min":1,"max":1},"_account_id":31542,"name":"Andrew Bonney","email":"andrew.bonney@bbc.co.uk","username":"andrewbonney"},{"value":0,"_account_id":28619,"name":"Dmitriy Rabotyagov","email":"noonedeadpunk@gmail.com","username":"noonedeadpunk"},{"value":0,"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]}],"values":{"-1":"Work in progress"," 0":"Ready for reviews","+1":"Approved"},"description":"","default_value":0,"optional":true},"Backport-Candidate":{"all":[{"value":0,"_account_id":31542,"name":"Andrew Bonney","email":"andrew.bonney@bbc.co.uk","username":"andrewbonney"},{"value":0,"_account_id":28619,"name":"Dmitriy Rabotyagov","email":"noonedeadpunk@gmail.com","username":"noonedeadpunk"},{"value":0,"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]}],"values":{"-1":"Not A Backport Candidate"," 0":"Backport Review Needed","+1":"Proposed Backport"},"description":"","default_value":0,"optional":true}},"removable_reviewers":[],"reviewers":{"REVIEWER":[{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},{"_account_id":28619,"name":"Dmitriy Rabotyagov","email":"noonedeadpunk@gmail.com","username":"noonedeadpunk"},{"_account_id":31542,"name":"Andrew Bonney","email":"andrew.bonney@bbc.co.uk","username":"andrewbonney"}]},"pending_reviewers":{},"reviewer_updates":[{"updated":"2023-09-28 09:09:04.000000000","updated_by":{"_account_id":28619,"name":"Dmitriy Rabotyagov","email":"noonedeadpunk@gmail.com","username":"noonedeadpunk"},"reviewer":{"_account_id":28619,"name":"Dmitriy Rabotyagov","email":"noonedeadpunk@gmail.com","username":"noonedeadpunk"},"state":"REVIEWER"},{"updated":"2023-09-28 10:59:35.000000000","updated_by":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"reviewer":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"state":"REVIEWER"},{"updated":"2023-10-02 14:30:32.000000000","updated_by":{"_account_id":31542,"name":"Andrew Bonney","email":"andrew.bonney@bbc.co.uk","username":"andrewbonney"},"reviewer":{"_account_id":31542,"name":"Andrew Bonney","email":"andrew.bonney@bbc.co.uk","username":"andrewbonney"},"state":"REVIEWER"}],"messages":[{"id":"9196e054ccd24e2a3fd0f4a4607581903a5f7cdc","tag":"autogenerated:gerrit:newPatchSet","author":{"_account_id":25023,"name":"Jonathan Rosser","email":"jonathan.rosser@rd.bbc.co.uk","username":"jrosser"},"date":"2023-09-28 09:00:16.000000000","message":"Uploaded patch set 1.","accounts_in_message":[],"_revision_number":1},{"id":"a9f341b1104d915604643c5465ab58b7405704d5","author":{"_account_id":28619,"name":"Dmitriy Rabotyagov","email":"noonedeadpunk@gmail.com","username":"noonedeadpunk"},"date":"2023-09-28 09:09:04.000000000","message":"Patch Set 1: Code-Review+2","accounts_in_message":[],"_revision_number":1},{"id":"bce81ddcd71947f9f4bd95331eed3ee6a9ca0400","tag":"autogenerated:zuul:check","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2023-09-28 10:59:35.000000000","message":"Patch Set 1: Verified-1\n\nBuild failed (check pipeline).  For information on how to proceed, see\nhttps://docs.opendev.org/opendev/infra-manual/latest/developers.html#automated-testing\n\nhttps://zuul.opendev.org/t/openstack/buildset/6403b27b54ba4b1da27e9ba0eb64cabf\n\n- openstack-ansible-linters-ubuntu-jammy https://zuul.opendev.org/t/openstack/build/a4d68230d4414396b17c57624574f8f2 : SUCCESS in 11m 08s\n- openstack-tox-docs https://zuul.opendev.org/t/openstack/build/5f6478e686ff436f9b27df8c8c4d1e8c : SUCCESS in 4m 53s\n- openstack-ansible-deploy-infra_lxc-centos-9-stream https://zuul.opendev.org/t/openstack/build/05d53c890c6c46d4be75e87a35a9603f : SUCCESS in 1h 47m 23s (non-voting)\n- openstack-ansible-deploy-infra_lxc-rockylinux-9 https://zuul.opendev.org/t/openstack/build/eb3c8c0f960f444ebd554d41153961a0 : SUCCESS in 1h 34m 22s\n- openstack-ansible-deploy-infra_lxc_tls-rockylinux-9 https://zuul.opendev.org/t/openstack/build/2f83c168d0744499a48c4145b50246a1 : FAILURE in 40m 23s\n- openstack-ansible-deploy-infra_lxc-debian-bullseye https://zuul.opendev.org/t/openstack/build/f4376f6f49fa4b41aa05f0b4e4f2500f : SUCCESS in 1h 11m 48s\n- openstack-ansible-deploy-infra_lxc-ubuntu-jammy https://zuul.opendev.org/t/openstack/build/235c2480bfde4dcda4476940c2126e33 : SUCCESS in 1h 16m 18s\n- openstack-ansible-deploy-infra_lxc_stepca-ubuntu-jammy https://zuul.opendev.org/t/openstack/build/fa0335ef0beb46e0b9d3d3aae96db0ba : SUCCESS in 1h 23m 51s\n- openstack-ansible-deploy-infra_lxc_tls-ubuntu-jammy https://zuul.opendev.org/t/openstack/build/0d67323f8f174a159d976be36efe0abe : SUCCESS in 1h 24m 27s\n- openstack-ansible-upgrade-infra_lxc-rockylinux-9 https://zuul.opendev.org/t/openstack/build/fa49b8bb2cdb4c3f850aa60cb8799107 : SUCCESS in 1h 54m 34s\n- openstack-ansible-upgrade-infra_lxc-ubuntu-jammy https://zuul.opendev.org/t/openstack/build/40208d02dad94fd79fbbf185d89a7db3 : SUCCESS in 1h 46m 36s\n- openstack-ansible-deploy-hosts_lxc_zfs-ubuntu-jammy https://zuul.opendev.org/t/openstack/build/57401c7e6e174c49844ce2420cd138f0 : SUCCESS in 28m 43s\n- openstack-ansible-deploy-hosts_lxc_btrfs-ubuntu-jammy https://zuul.opendev.org/t/openstack/build/10a7ca83fbb14fa0bcf26f41f3231398 : SUCCESS in 28m 23s\n- openstack-ansible-deploy-hosts_lxc_overlayfs-ubuntu-jammy https://zuul.opendev.org/t/openstack/build/6d9d1ccb514b4f8184b2dff5a58a2ce1 : SUCCESS in 33m 55s\n- openstack-ansible-deploy-hosts_lxc_btrfs-debian-bullseye https://zuul.opendev.org/t/openstack/build/74254bf109774523aec8bacfc7939f8d : SUCCESS in 22m 21s\n- openstack-ansible-deploy-hosts_lxc_overlayfs-debian-bullseye https://zuul.opendev.org/t/openstack/build/2c8e1eff2901460bb21d5a8474475294 : SUCCESS in 29m 15s\n- openstack-ansible-deploy-hosts_lxc_overlayfs-rockylinux-9 https://zuul.opendev.org/t/openstack/build/bb6269354ba34fbd902c5fdfa9c12af9 : FAILURE in 33m 05s","accounts_in_message":[],"_revision_number":1},{"id":"1730867033908aa0688a47a0d4b297e4cb1a26ef","author":{"_account_id":28619,"name":"Dmitriy Rabotyagov","email":"noonedeadpunk@gmail.com","username":"noonedeadpunk"},"date":"2023-09-29 08:10:23.000000000","message":"Patch Set 1:\n\n(1 comment)","accounts_in_message":[],"_revision_number":1},{"id":"f02fc4e6437a0fd9a8e2b6afd766dbb10e722036","tag":"autogenerated:zuul:check","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2023-09-29 10:28:18.000000000","message":"Patch Set 1: Verified+1\n\nBuild succeeded (check pipeline).\nhttps://zuul.opendev.org/t/openstack/buildset/b8f5ac87b51c459b9ba24be2aa3eb88b\n\n- openstack-ansible-linters-ubuntu-jammy https://zuul.opendev.org/t/openstack/build/1aaf626b285f4588b71e22027280f9b0 : SUCCESS in 12m 40s\n- openstack-tox-docs https://zuul.opendev.org/t/openstack/build/73c365e976024db2bfc1f50bd8468eb6 : SUCCESS in 4m 14s\n- openstack-ansible-deploy-infra_lxc-centos-9-stream https://zuul.opendev.org/t/openstack/build/5431e5c0595e447ab76738a4a35fce6a : SUCCESS in 1h 25m 36s (non-voting)\n- openstack-ansible-deploy-infra_lxc-rockylinux-9 https://zuul.opendev.org/t/openstack/build/1cdba32e94a749f38ad1a3525a02918d : SUCCESS in 1h 24m 02s\n- openstack-ansible-deploy-infra_lxc_tls-rockylinux-9 https://zuul.opendev.org/t/openstack/build/094a428ca2e24544b94ef50cd0d730c4 : SUCCESS in 1h 24m 45s\n- openstack-ansible-deploy-infra_lxc-debian-bullseye https://zuul.opendev.org/t/openstack/build/942d1cd649e940e09a2d1b30cc1267b7 : SUCCESS in 1h 17m 05s\n- openstack-ansible-deploy-infra_lxc-ubuntu-jammy https://zuul.opendev.org/t/openstack/build/6d3dfd60ad804a0ab1ee66789bc4ae84 : SUCCESS in 1h 22m 37s\n- openstack-ansible-deploy-infra_lxc_stepca-ubuntu-jammy https://zuul.opendev.org/t/openstack/build/b43641847d1e45a8b04e1d040b41edad : SUCCESS in 1h 19m 07s\n- openstack-ansible-deploy-infra_lxc_tls-ubuntu-jammy https://zuul.opendev.org/t/openstack/build/2a31fbae6b5f4bd1a29443a1c93b1cda : SUCCESS in 1h 15m 47s\n- openstack-ansible-upgrade-infra_lxc-rockylinux-9 https://zuul.opendev.org/t/openstack/build/90355e7082564238b33d805aa31d376f : SUCCESS in 2h 07m 51s\n- openstack-ansible-upgrade-infra_lxc-ubuntu-jammy https://zuul.opendev.org/t/openstack/build/5fa3c5cc2b4e464e9a4de063571a0da9 : SUCCESS in 1h 44m 36s\n- openstack-ansible-deploy-hosts_lxc_zfs-ubuntu-jammy https://zuul.opendev.org/t/openstack/build/7207612c11fd463fb3e566feac78b31b : SUCCESS in 29m 57s\n- openstack-ansible-deploy-hosts_lxc_btrfs-ubuntu-jammy https://zuul.opendev.org/t/openstack/build/30c9f567dd624c5bbc41648c7a83b5e9 : SUCCESS in 34m 41s\n- openstack-ansible-deploy-hosts_lxc_overlayfs-ubuntu-jammy https://zuul.opendev.org/t/openstack/build/5ed50e1e21eb40e18df90988f4679488 : SUCCESS in 33m 39s\n- openstack-ansible-deploy-hosts_lxc_btrfs-debian-bullseye https://zuul.opendev.org/t/openstack/build/cbc44f56680d4dd29c89a687252ed43f : SUCCESS in 24m 13s\n- openstack-ansible-deploy-hosts_lxc_overlayfs-debian-bullseye https://zuul.opendev.org/t/openstack/build/1c0169d0ca5e459e854577b2ef9b9b5c : SUCCESS in 36m 25s\n- openstack-ansible-deploy-hosts_lxc_overlayfs-rockylinux-9 https://zuul.opendev.org/t/openstack/build/53d04792f6e24b82927415ed3932e502 : SUCCESS in 31m 40s","accounts_in_message":[],"_revision_number":1},{"id":"829b3cd09784495ac219d8bdb0d6cab9b448c604","author":{"_account_id":31542,"name":"Andrew Bonney","email":"andrew.bonney@bbc.co.uk","username":"andrewbonney"},"date":"2023-10-02 14:30:32.000000000","message":"Patch Set 1: Code-Review+2 Workflow+1","accounts_in_message":[],"_revision_number":1},{"id":"4500d7aba735b8c5445707f6b05a7caee3b22dbb","tag":"autogenerated:zuul:gate","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2023-10-02 14:32:49.000000000","message":"Patch Set 1: -Verified\n\nStarting gate jobs.","accounts_in_message":[],"_revision_number":1},{"id":"d66ad5ec55e436fea15e1319055162c98c36cab7","tag":"autogenerated:zuul:gate","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2023-10-02 16:21:21.000000000","message":"Patch Set 1: Verified-2\n\nBuild failed (gate pipeline).  For information on how to proceed, see\nhttps://docs.opendev.org/opendev/infra-manual/latest/developers.html#automated-testing\n\nhttps://zuul.opendev.org/t/openstack/buildset/fe26e41525894b06bcb600f9a5d1242d\n\n- openstack-ansible-linters-ubuntu-jammy https://zuul.opendev.org/t/openstack/build/fc221f573ff8427a882d0684387530a7 : SUCCESS in 13m 07s\n- openstack-tox-docs https://zuul.opendev.org/t/openstack/build/66d08ccc9bf244aa8eb8c0a171b99377 : SUCCESS in 4m 33s\n- openstack-ansible-deploy-infra_lxc-rockylinux-9 https://zuul.opendev.org/t/openstack/build/28eba405e8c24ded8d879cbf192f091b : SUCCESS in 1h 34m 42s\n- openstack-ansible-deploy-infra_lxc_tls-rockylinux-9 https://zuul.opendev.org/t/openstack/build/0a97f17f3f9643a981af09a443c340c2 : SUCCESS in 1h 47m 14s\n- openstack-ansible-deploy-infra_lxc-debian-bullseye https://zuul.opendev.org/t/openstack/build/d3ee8a7f9517418a8a0a35a3d0dae6cb : SUCCESS in 1h 16m 58s\n- openstack-ansible-deploy-infra_lxc-ubuntu-jammy https://zuul.opendev.org/t/openstack/build/df72e3979d8747ceac0a72a08f249371 : SUCCESS in 1h 23m 43s\n- openstack-ansible-deploy-infra_lxc_stepca-ubuntu-jammy https://zuul.opendev.org/t/openstack/build/76da4e212f164f0abc4bf6c8bf23135c : SUCCESS in 1h 24m 04s\n- openstack-ansible-deploy-infra_lxc_tls-ubuntu-jammy https://zuul.opendev.org/t/openstack/build/d75f56017a744410a390df0b0644a045 : SUCCESS in 1h 31m 53s\n- openstack-ansible-upgrade-infra_lxc-rockylinux-9 https://zuul.opendev.org/t/openstack/build/83d19074407c4a9ab0880fdf0c0a0185 : FAILURE in 24m 49s\n- openstack-ansible-upgrade-infra_lxc-ubuntu-jammy https://zuul.opendev.org/t/openstack/build/2e7190a62ae542aa8105efbe4c4c93f8 : SUCCESS in 1h 39m 13s\n- openstack-ansible-deploy-hosts_lxc_zfs-ubuntu-jammy https://zuul.opendev.org/t/openstack/build/c9d5b2df767b4c359939802aa8b80665 : SUCCESS in 28m 59s\n- openstack-ansible-deploy-hosts_lxc_btrfs-ubuntu-jammy https://zuul.opendev.org/t/openstack/build/03f2172547b541cd9d9c91e10994f2bd : SUCCESS in 32m 37s\n- openstack-ansible-deploy-hosts_lxc_overlayfs-ubuntu-jammy https://zuul.opendev.org/t/openstack/build/2ec7c31733ae4566847355e43efbc905 : SUCCESS in 27m 42s\n- openstack-ansible-deploy-hosts_lxc_btrfs-debian-bullseye https://zuul.opendev.org/t/openstack/build/92588145fa59433ca4bd3457539c5473 : SUCCESS in 27m 34s\n- openstack-ansible-deploy-hosts_lxc_overlayfs-debian-bullseye https://zuul.opendev.org/t/openstack/build/2e02b16360a940e0aa9a7ff74861f1c1 : SUCCESS in 25m 43s\n- openstack-ansible-deploy-hosts_lxc_overlayfs-rockylinux-9 https://zuul.opendev.org/t/openstack/build/7d817584d10b4f9488243e4bd252a0d8 : SUCCESS in 25m 08s","accounts_in_message":[],"_revision_number":1},{"id":"b1234200f6980eb612d29142c46097c4674a87eb","author":{"_account_id":25023,"name":"Jonathan Rosser","email":"jonathan.rosser@rd.bbc.co.uk","username":"jrosser"},"date":"2023-10-02 16:42:03.000000000","message":"Patch Set 1:\n\n(1 comment)","accounts_in_message":[],"_revision_number":1},{"id":"2cf71beb8f4237723de00547100923b734a98ffe","tag":"autogenerated:zuul:check","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2023-10-02 18:44:03.000000000","message":"Patch Set 1: Verified+1\n\nBuild succeeded (check pipeline).\nhttps://zuul.opendev.org/t/openstack/buildset/175543a8ff354da8b04ad781ef0aeed1\n\n- openstack-ansible-linters-ubuntu-jammy https://zuul.opendev.org/t/openstack/build/bbd03bff81a1409dac529cf3f1921420 : SUCCESS in 17m 03s\n- openstack-tox-docs https://zuul.opendev.org/t/openstack/build/5ae70dc1563f46539bbe42aaa8b5e834 : SUCCESS in 5m 26s\n- openstack-ansible-deploy-infra_lxc-centos-9-stream https://zuul.opendev.org/t/openstack/build/3bc8ad74cfaa42c18e7b3a4e493d0355 : SUCCESS in 1h 33m 26s (non-voting)\n- openstack-ansible-deploy-infra_lxc-rockylinux-9 https://zuul.opendev.org/t/openstack/build/9544bf31a6e14a8cadecb51a1b9f2c22 : SUCCESS in 1h 31m 23s\n- openstack-ansible-deploy-infra_lxc_tls-rockylinux-9 https://zuul.opendev.org/t/openstack/build/afdf9e1887c141488ad58fcd8884dfa8 : SUCCESS in 1h 34m 12s\n- openstack-ansible-deploy-infra_lxc-debian-bullseye https://zuul.opendev.org/t/openstack/build/5ac4a1a5e3c24b6e97eae542b010fd83 : SUCCESS in 1h 14m 26s\n- openstack-ansible-deploy-infra_lxc-ubuntu-jammy https://zuul.opendev.org/t/openstack/build/0443206918d64ec680237ae2b6950ee9 : SUCCESS in 1h 23m 37s\n- openstack-ansible-deploy-infra_lxc_stepca-ubuntu-jammy https://zuul.opendev.org/t/openstack/build/b9945a6a446e48febe3d446511d651b8 : SUCCESS in 1h 16m 57s\n- openstack-ansible-deploy-infra_lxc_tls-ubuntu-jammy https://zuul.opendev.org/t/openstack/build/66d986bade964c00a0e405ce97b27ef8 : SUCCESS in 1h 22m 13s\n- openstack-ansible-upgrade-infra_lxc-rockylinux-9 https://zuul.opendev.org/t/openstack/build/adcdd290cbbb4c928dd4a2c82abf7ddc : SUCCESS in 1h 43m 23s\n- openstack-ansible-upgrade-infra_lxc-ubuntu-jammy https://zuul.opendev.org/t/openstack/build/710093488ee84f0681f8c5af469e4532 : SUCCESS in 1h 52m 47s\n- openstack-ansible-deploy-hosts_lxc_zfs-ubuntu-jammy https://zuul.opendev.org/t/openstack/build/8f6f2c3e09684a089a93dabd8ca79f57 : SUCCESS in 28m 53s\n- openstack-ansible-deploy-hosts_lxc_btrfs-ubuntu-jammy https://zuul.opendev.org/t/openstack/build/b3a59132aa91410ea91c9b6cf2e8abb8 : SUCCESS in 26m 31s\n- openstack-ansible-deploy-hosts_lxc_overlayfs-ubuntu-jammy https://zuul.opendev.org/t/openstack/build/3daaac0fe44642bca8c02ae60035f0fa : SUCCESS in 25m 46s\n- openstack-ansible-deploy-hosts_lxc_btrfs-debian-bullseye https://zuul.opendev.org/t/openstack/build/a9e716c0bcef4f79a5487332f2b7a718 : SUCCESS in 32m 47s\n- openstack-ansible-deploy-hosts_lxc_overlayfs-debian-bullseye https://zuul.opendev.org/t/openstack/build/06608e5898864cb788ea7504f68055f8 : SUCCESS in 28m 51s\n- openstack-ansible-deploy-hosts_lxc_overlayfs-rockylinux-9 https://zuul.opendev.org/t/openstack/build/fa7a5864a39c4ff9ae47f794e23ecaa2 : SUCCESS in 27m 28s","accounts_in_message":[],"_revision_number":1},{"id":"58d48eb73d67bd5521b63a2ac1c920557c89183a","tag":"autogenerated:zuul:gate","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2023-10-02 18:46:13.000000000","message":"Patch Set 1: -Verified\n\nStarting gate jobs.","accounts_in_message":[],"_revision_number":1},{"id":"5dbaedbe39f7f5abfd6c9f8ee95471fe0e9e1d3b","tag":"autogenerated:zuul:gate","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2023-10-02 20:52:38.000000000","message":"Patch Set 1: Verified+2\n\nBuild succeeded (gate pipeline).\nhttps://zuul.opendev.org/t/openstack/buildset/96fbb91eee3c411eb5c50e1bddaab1d2\n\n- openstack-ansible-linters-ubuntu-jammy https://zuul.opendev.org/t/openstack/build/b6f5410e10664c9990c146c2c3c98a59 : SUCCESS in 17m 01s\n- openstack-tox-docs https://zuul.opendev.org/t/openstack/build/ef46836a5f7d474eb9382ed3e34def77 : SUCCESS in 4m 43s\n- openstack-ansible-deploy-infra_lxc-rockylinux-9 https://zuul.opendev.org/t/openstack/build/3d6adc30b084473cadd54ac53eb786c9 : SUCCESS in 1h 25m 20s\n- openstack-ansible-deploy-infra_lxc_tls-rockylinux-9 https://zuul.opendev.org/t/openstack/build/a101021c83b547d68718dbbd1c03e779 : SUCCESS in 1h 46m 11s\n- openstack-ansible-deploy-infra_lxc-debian-bullseye https://zuul.opendev.org/t/openstack/build/d6b851019cef44d88a242c3ae8228238 : SUCCESS in 1h 15m 50s\n- openstack-ansible-deploy-infra_lxc-ubuntu-jammy https://zuul.opendev.org/t/openstack/build/e9c251002de04bed867b6ad94e10a021 : SUCCESS in 1h 27m 10s\n- openstack-ansible-deploy-infra_lxc_stepca-ubuntu-jammy https://zuul.opendev.org/t/openstack/build/8c4c128475d347d58be9207321e6ba84 : SUCCESS in 1h 20m 11s\n- openstack-ansible-deploy-infra_lxc_tls-ubuntu-jammy https://zuul.opendev.org/t/openstack/build/b94b0cad27e14229a47ee9f0bca09368 : SUCCESS in 1h 25m 29s\n- openstack-ansible-upgrade-infra_lxc-rockylinux-9 https://zuul.opendev.org/t/openstack/build/4b17cb1c517d46398624fefca54e03dc : SUCCESS in 2h 02m 25s\n- openstack-ansible-upgrade-infra_lxc-ubuntu-jammy https://zuul.opendev.org/t/openstack/build/b39d17320cd64de9ab8da3fd0992b386 : SUCCESS in 1h 44m 02s\n- openstack-ansible-deploy-hosts_lxc_zfs-ubuntu-jammy https://zuul.opendev.org/t/openstack/build/379b4e9a44d1438b87d87f6a73d73610 : SUCCESS in 29m 13s\n- openstack-ansible-deploy-hosts_lxc_btrfs-ubuntu-jammy https://zuul.opendev.org/t/openstack/build/c27f6b14e69245519d6bd249de51aee7 : SUCCESS in 26m 09s\n- openstack-ansible-deploy-hosts_lxc_overlayfs-ubuntu-jammy https://zuul.opendev.org/t/openstack/build/912728cd239047138ed56530dcac2c94 : SUCCESS in 26m 37s\n- openstack-ansible-deploy-hosts_lxc_btrfs-debian-bullseye https://zuul.opendev.org/t/openstack/build/917fbb151f414aeb9c270b8f79c01266 : SUCCESS in 24m 25s\n- openstack-ansible-deploy-hosts_lxc_overlayfs-debian-bullseye https://zuul.opendev.org/t/openstack/build/5d9d2151ecd6403e83740948ea634883 : SUCCESS in 27m 42s\n- openstack-ansible-deploy-hosts_lxc_overlayfs-rockylinux-9 https://zuul.opendev.org/t/openstack/build/cbbfabd91c6743ab89668a6daa3e89ff : SUCCESS in 30m 05s","accounts_in_message":[],"_revision_number":1},{"id":"a14242b08d57473e401ba748cf90572f936343ff","tag":"autogenerated:gerrit:merged","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2023-10-02 20:52:38.000000000","message":"Change has been successfully merged","accounts_in_message":[],"_revision_number":1},{"id":"42adfb96a4d091e8c396c0edd91439c35e54da08","tag":"autogenerated:zuul:promote","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2023-10-02 20:53:36.000000000","message":"Patch Set 1:\n\nBuild succeeded (promote pipeline).\nhttps://zuul.opendev.org/t/openstack/buildset/98c7e4ab191e451f9bf30e848d6f2f2f\n\n- promote-openstack-tox-docs https://zuul.opendev.org/t/openstack/build/777b840fbabd4f38a0807f6c77816cf1 : SUCCESS in 41s","accounts_in_message":[],"_revision_number":1}],"current_revision_number":1,"current_revision":"42cfa88bb56f1cb29bb07f499a58ff9d6631f6cc","revisions":{"42cfa88bb56f1cb29bb07f499a58ff9d6631f6cc":{"kind":"REWORK","_number":1,"created":"2023-09-28 09:00:16.000000000","uploader":{"_account_id":25023,"name":"Jonathan Rosser","email":"jonathan.rosser@rd.bbc.co.uk","username":"jrosser"},"ref":"refs/changes/50/896750/1","fetch":{"anonymous http":{"url":"https://review.opendev.org/openstack/openstack-ansible-lxc_hosts","ref":"refs/changes/50/896750/1","commands":{"Checkout":"git fetch https://review.opendev.org/openstack/openstack-ansible-lxc_hosts refs/changes/50/896750/1 \u0026\u0026 git checkout FETCH_HEAD","Cherry Pick":"git fetch https://review.opendev.org/openstack/openstack-ansible-lxc_hosts refs/changes/50/896750/1 \u0026\u0026 git cherry-pick FETCH_HEAD","Format Patch":"git fetch https://review.opendev.org/openstack/openstack-ansible-lxc_hosts refs/changes/50/896750/1 \u0026\u0026 git format-patch -1 --stdout FETCH_HEAD","Pull":"git pull https://review.opendev.org/openstack/openstack-ansible-lxc_hosts refs/changes/50/896750/1"}}},"commit":{"parents":[{"commit":"2272de8f0c5977819ed38582c30cff68d6da7d5b","subject":"Fix linters issues","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/openstack-ansible-lxc_hosts/commit/2272de8f0c5977819ed38582c30cff68d6da7d5b"}]}],"author":{"name":"Jonathan Rosser","email":"jonathan.rosser@rd.bbc.co.uk","date":"2023-09-28 08:59:52.000000000","tz":60},"committer":{"name":"Jonathan Rosser","email":"jonathan.rosser@rd.bbc.co.uk","date":"2023-09-28 09:00:13.000000000","tz":60},"subject":"Fix linter error","message":"Fix linter error\n\nSplit long line\n\nChange-Id: I2466a9959bc93da754e11d8b9c6fb5d90f64163f\n","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/openstack-ansible-lxc_hosts/commit/42cfa88bb56f1cb29bb07f499a58ff9d6631f6cc"}],"resolve_conflicts_web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/openstack-ansible-lxc_hosts/commit/42cfa88bb56f1cb29bb07f499a58ff9d6631f6cc"}]},"branch":"refs/heads/master"}},"requirements":[],"submit_records":[{"rule_name":"gerrit~DefaultSubmitRule","status":"CLOSED","labels":[{"label":"Verified","status":"MAY","applied_by":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]}},{"label":"Code-Review","status":"MAY","applied_by":{"_account_id":31542,"name":"Andrew Bonney","email":"andrew.bonney@bbc.co.uk","username":"andrewbonney"}},{"label":"Workflow","status":"MAY","applied_by":{"_account_id":31542,"name":"Andrew Bonney","email":"andrew.bonney@bbc.co.uk","username":"andrewbonney"}},{"label":"Backport-Candidate","status":"MAY"}]}],"submit_requirements":[{"name":"Verified","description":"Verified in gate by CI","status":"SATISFIED","is_legacy":false,"submittability_expression_result":{"expression":"label:Verified\u003dMAX AND -label:Verified\u003dMIN","fulfilled":true,"status":"PASS","passing_atoms":["label:Verified\u003dMAX"],"failing_atoms":["label:Verified\u003dMIN"],"atom_explanations":{}}},{"name":"Backport-Candidate","description":"Backport candidate status","status":"NOT_APPLICABLE","is_legacy":false,"applicability_expression_result":{"fulfilled":false,"status":"FAIL"},"submittability_expression_result":{"expression":"is:true","fulfilled":true,"status":"NOT_EVALUATED","passing_atoms":[],"failing_atoms":[],"atom_explanations":{}}},{"name":"Code-Review","description":"Code reviewed by core reviewer","status":"SATISFIED","is_legacy":false,"submittability_expression_result":{"expression":"label:Code-Review\u003dMAX AND -label:Code-Review\u003dMIN","fulfilled":true,"status":"PASS","passing_atoms":["label:Code-Review\u003dMAX"],"failing_atoms":["label:Code-Review\u003dMIN"],"atom_explanations":{}}},{"name":"Workflow","description":"Approved for gate by core reviewer","status":"SATISFIED","is_legacy":false,"submittability_expression_result":{"expression":"label:Workflow\u003dMAX AND -label:Workflow\u003dMIN","fulfilled":true,"status":"PASS","passing_atoms":["label:Workflow\u003dMAX"],"failing_atoms":["label:Workflow\u003dMIN"],"atom_explanations":{}}}]}
