)]}'
{"id":"openstack%2Fopenstack-ansible-lxc_hosts~897861","triplet_id":"openstack%2Fopenstack-ansible-lxc_hosts~master~I1d2be9d07b38eaf2b737819c451a0d2339f723d0","project":"openstack/openstack-ansible-lxc_hosts","branch":"master","attention_set":{},"removed_from_attention_set":{"25023":{"account":{"_account_id":25023,"name":"Jonathan Rosser","email":"jonathan.rosser@rd.bbc.co.uk","username":"jrosser"},"last_update":"2023-10-12 05:03:32.000000000","reason":"Change was submitted"}},"hashtags":[],"change_id":"I1d2be9d07b38eaf2b737819c451a0d2339f723d0","subject":"Remove lxc_cache_map variable","status":"MERGED","created":"2023-10-10 16:51:42.000000000","updated":"2023-10-12 05:05:16.000000000","submitted":"2023-10-12 05:03:32.000000000","submitter":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"total_comment_count":1,"unresolved_comment_count":1,"has_review_started":true,"submission_id":"897861","meta_rev_id":"893c89197f0a342aeb3c13c1c392cec61c598173","_number":897861,"virtual_id_number":897861,"owner":{"_account_id":25023,"name":"Jonathan Rosser","email":"jonathan.rosser@rd.bbc.co.uk","username":"jrosser"},"actions":{},"labels":{"Verified":{"approved":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"all":[{"value":0,"_account_id":28619,"name":"Dmitriy Rabotyagov","email":"noonedeadpunk@gmail.com","username":"noonedeadpunk"},{"value":0,"_account_id":34411,"name":"Neil Hanlon","email":"neil@shrug.pw","username":"nhanlon"},{"tag":"autogenerated:zuul:gate","value":2,"date":"2023-10-12 05:03:32.000000000","permitted_voting_range":{"min":2,"max":2},"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]}],"values":{"-2":"Fails","-1":"Doesn\u0027t seem to work"," 0":"No score","+1":"Works for me","+2":"Verified"},"description":"","default_value":0,"optional":true},"Code-Review":{"approved":{"_account_id":28619,"name":"Dmitriy Rabotyagov","email":"noonedeadpunk@gmail.com","username":"noonedeadpunk"},"all":[{"value":2,"date":"2023-10-11 18:17:04.000000000","permitted_voting_range":{"min":2,"max":2},"_account_id":28619,"name":"Dmitriy Rabotyagov","email":"noonedeadpunk@gmail.com","username":"noonedeadpunk"},{"value":2,"date":"2023-10-12 02:58:34.000000000","permitted_voting_range":{"min":2,"max":2},"_account_id":34411,"name":"Neil Hanlon","email":"neil@shrug.pw","username":"nhanlon"},{"value":0,"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]}],"values":{"-2":"Do not merge","-1":"This patch needs further work before it can be merged"," 0":"No score","+1":"Looks good to me, but someone else must approve","+2":"Looks good to me (core reviewer)"},"description":"","default_value":0,"optional":true},"Workflow":{"approved":{"_account_id":34411,"name":"Neil Hanlon","email":"neil@shrug.pw","username":"nhanlon"},"all":[{"value":0,"_account_id":28619,"name":"Dmitriy Rabotyagov","email":"noonedeadpunk@gmail.com","username":"noonedeadpunk"},{"value":1,"date":"2023-10-12 02:58:34.000000000","permitted_voting_range":{"min":1,"max":1},"_account_id":34411,"name":"Neil Hanlon","email":"neil@shrug.pw","username":"nhanlon"},{"value":0,"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]}],"values":{"-1":"Work in progress"," 0":"Ready for reviews","+1":"Approved"},"description":"","default_value":0,"optional":true},"Backport-Candidate":{"all":[{"value":0,"_account_id":28619,"name":"Dmitriy Rabotyagov","email":"noonedeadpunk@gmail.com","username":"noonedeadpunk"},{"value":0,"_account_id":34411,"name":"Neil Hanlon","email":"neil@shrug.pw","username":"nhanlon"},{"value":0,"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]}],"values":{"-1":"Not A Backport Candidate"," 0":"Backport Review Needed","+1":"Proposed Backport"},"description":"","default_value":0,"optional":true}},"removable_reviewers":[],"reviewers":{"REVIEWER":[{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},{"_account_id":28619,"name":"Dmitriy Rabotyagov","email":"noonedeadpunk@gmail.com","username":"noonedeadpunk"},{"_account_id":34411,"name":"Neil Hanlon","email":"neil@shrug.pw","username":"nhanlon"}]},"pending_reviewers":{},"reviewer_updates":[{"updated":"2023-10-10 18:49:21.000000000","updated_by":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"reviewer":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"state":"REVIEWER"},{"updated":"2023-10-11 10:44:45.000000000","updated_by":{"_account_id":28619,"name":"Dmitriy Rabotyagov","email":"noonedeadpunk@gmail.com","username":"noonedeadpunk"},"reviewer":{"_account_id":28619,"name":"Dmitriy Rabotyagov","email":"noonedeadpunk@gmail.com","username":"noonedeadpunk"},"state":"CC"},{"updated":"2023-10-11 18:17:04.000000000","updated_by":{"_account_id":28619,"name":"Dmitriy Rabotyagov","email":"noonedeadpunk@gmail.com","username":"noonedeadpunk"},"reviewer":{"_account_id":28619,"name":"Dmitriy Rabotyagov","email":"noonedeadpunk@gmail.com","username":"noonedeadpunk"},"state":"REVIEWER"},{"updated":"2023-10-12 02:58:34.000000000","updated_by":{"_account_id":34411,"name":"Neil Hanlon","email":"neil@shrug.pw","username":"nhanlon"},"reviewer":{"_account_id":34411,"name":"Neil Hanlon","email":"neil@shrug.pw","username":"nhanlon"},"state":"REVIEWER"}],"messages":[{"id":"ebc1b8d28b81a5c30eaa004422e553a9adaf228c","tag":"autogenerated:gerrit:newPatchSet","author":{"_account_id":25023,"name":"Jonathan Rosser","email":"jonathan.rosser@rd.bbc.co.uk","username":"jrosser"},"date":"2023-10-10 16:51:42.000000000","message":"Uploaded patch set 1.","accounts_in_message":[],"_revision_number":1},{"id":"c148f5a4372e91f76126427db777aabb573c7938","tag":"autogenerated:zuul:check","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2023-10-10 18:49:21.000000000","message":"Patch Set 1: Verified-1\n\nBuild failed (check pipeline).  For information on how to proceed, see\nhttps://docs.opendev.org/opendev/infra-manual/latest/developers.html#automated-testing\n\nhttps://zuul.opendev.org/t/openstack/buildset/9b42a5d2db21486898af23df87bce8ad\n\n- openstack-ansible-linters-ubuntu-jammy https://zuul.opendev.org/t/openstack/build/4647dca91f3d443bbc8ec239c9c0e7a4 : FAILURE in 16m 46s\n- openstack-tox-docs https://zuul.opendev.org/t/openstack/build/731289470eab4bb480d05dc6ead2aede : SUCCESS in 4m 45s\n- build-openstack-releasenotes https://zuul.opendev.org/t/openstack/build/1314083ae19941d7aff1b0e2a31dc30c : SUCCESS in 4m 11s\n- openstack-ansible-deploy-infra_lxc-centos-9-stream https://zuul.opendev.org/t/openstack/build/f0582ae20d284d1cb4462322aba6c98c : SUCCESS in 1h 30m 13s (non-voting)\n- openstack-ansible-deploy-infra_lxc-rockylinux-9 https://zuul.opendev.org/t/openstack/build/c497901d340a4e5e98a2d5e2c8f77b2b : SUCCESS in 1h 32m 15s\n- openstack-ansible-deploy-infra_lxc_tls-rockylinux-9 https://zuul.opendev.org/t/openstack/build/ce621772dcec4b27b6302072e25412b6 : SUCCESS in 1h 34m 41s\n- openstack-ansible-deploy-infra_lxc-debian-bullseye https://zuul.opendev.org/t/openstack/build/1c580a17d31d4cff962c696f6706b26c : SUCCESS in 1h 09m 49s\n- openstack-ansible-deploy-infra_lxc-ubuntu-jammy https://zuul.opendev.org/t/openstack/build/949c495119bf43939c446cf693fdc5a0 : SUCCESS in 1h 24m 08s\n- openstack-ansible-deploy-infra_lxc_stepca-ubuntu-jammy https://zuul.opendev.org/t/openstack/build/644c6182288e4225a3887579ecc8179b : SUCCESS in 1h 18m 46s\n- openstack-ansible-deploy-infra_lxc_tls-ubuntu-jammy https://zuul.opendev.org/t/openstack/build/3143a15ac4c94a90b1d3f936f375a83b : SUCCESS in 1h 13m 42s\n- openstack-ansible-upgrade-infra_lxc-rockylinux-9 https://zuul.opendev.org/t/openstack/build/76e7fc4522ea44ca847ffb6f22cc5760 : SUCCESS in 1h 54m 18s\n- openstack-ansible-upgrade-infra_lxc-ubuntu-jammy https://zuul.opendev.org/t/openstack/build/1752ed9582804ccc95edaccc9b8188f6 : SUCCESS in 1h 46m 26s\n- openstack-ansible-deploy-hosts_lxc_zfs-ubuntu-jammy https://zuul.opendev.org/t/openstack/build/4299076a4d494b08a600176615c0f469 : SUCCESS in 30m 36s\n- openstack-ansible-deploy-hosts_lxc_btrfs-ubuntu-jammy https://zuul.opendev.org/t/openstack/build/b6541531dfff490cb84aaf8eff29ef17 : SUCCESS in 27m 57s\n- openstack-ansible-deploy-hosts_lxc_overlayfs-ubuntu-jammy https://zuul.opendev.org/t/openstack/build/58ebf5d67e854db684dfc3dfd84a8757 : SUCCESS in 27m 12s\n- openstack-ansible-deploy-hosts_lxc_btrfs-debian-bullseye https://zuul.opendev.org/t/openstack/build/8c0c9333df9b47d3b7ba300438b42272 : SUCCESS in 29m 18s\n- openstack-ansible-deploy-hosts_lxc_overlayfs-debian-bullseye https://zuul.opendev.org/t/openstack/build/c9b5b576f5734eb39673b7488c7ee6a0 : SUCCESS in 29m 54s\n- openstack-ansible-deploy-hosts_lxc_overlayfs-rockylinux-9 https://zuul.opendev.org/t/openstack/build/f877b89a198049a49790f79f978ff9c8 : SUCCESS in 29m 56s","accounts_in_message":[],"_revision_number":1},{"id":"00abb8593edc977edcf72580a863ca076763c670","author":{"_account_id":28619,"name":"Dmitriy Rabotyagov","email":"noonedeadpunk@gmail.com","username":"noonedeadpunk"},"date":"2023-10-11 10:44:45.000000000","message":"Patch Set 1:\n\n(1 comment)","accounts_in_message":[],"_revision_number":1},{"id":"6939d2eca51313d1c7ede1dbde66edad52d765bd","tag":"autogenerated:gerrit:newPatchSet","author":{"_account_id":25023,"name":"Jonathan Rosser","email":"jonathan.rosser@rd.bbc.co.uk","username":"jrosser"},"date":"2023-10-11 13:22:26.000000000","message":"Patch Set 2: Published edit on patch set 1.\n\nOutdated Votes:\n* Verified-1\n","accounts_in_message":[],"_revision_number":2},{"id":"0acfae70573b6d1d1b0c930df415b24c1cd2bf60","tag":"autogenerated:zuul:check","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2023-10-11 14:39:16.000000000","message":"Patch Set 2: Verified-1\n\nBuild failed (check pipeline).  For information on how to proceed, see\nhttps://docs.opendev.org/opendev/infra-manual/latest/developers.html#automated-testing\n\nhttps://zuul.opendev.org/t/openstack/buildset/3d220005434c4a99b91c4e11506d3f9a\n\n- openstack-ansible-linters-ubuntu-jammy https://zuul.opendev.org/t/openstack/build/48c89107a90f47458dd3981d29c3b3ff : SUCCESS in 14m 58s\n- openstack-tox-docs https://zuul.opendev.org/t/openstack/build/671afd53355a44e095bd3647d1619992 : SUCCESS in 5m 22s\n- build-openstack-releasenotes https://zuul.opendev.org/t/openstack/build/9529262417bd4fc4b6008eb182a48690 : SUCCESS in 3m 19s\n- openstack-ansible-deploy-infra_lxc-centos-9-stream https://zuul.opendev.org/t/openstack/build/25e1c72aba81496aad439b57d3b30e8a : FAILURE in 23m 04s (non-voting)\n- openstack-ansible-deploy-infra_lxc-rockylinux-9 https://zuul.opendev.org/t/openstack/build/a864d2270e7b402093e18ccc58ae4543 : FAILURE in 25m 18s\n- openstack-ansible-deploy-infra_lxc_tls-rockylinux-9 https://zuul.opendev.org/t/openstack/build/1f6cc86f20a34eb8ac4b1153d42d4878 : FAILURE in 23m 55s\n- openstack-ansible-deploy-infra_lxc-debian-bullseye https://zuul.opendev.org/t/openstack/build/4154ce3e781c4839b9fabea99c8fb697 : FAILURE in 20m 41s\n- openstack-ansible-deploy-infra_lxc-ubuntu-jammy https://zuul.opendev.org/t/openstack/build/42514e9f2bbb48c78c3d97904e7b9cba : FAILURE in 14m 37s\n- openstack-ansible-deploy-infra_lxc_stepca-ubuntu-jammy https://zuul.opendev.org/t/openstack/build/011b7b2e1a1641fdb771fba31990ec9a : FAILURE in 16m 25s\n- openstack-ansible-deploy-infra_lxc_tls-ubuntu-jammy https://zuul.opendev.org/t/openstack/build/85375b76462140409c1c5c88e312736f : FAILURE in 19m 44s\n- openstack-ansible-upgrade-infra_lxc-rockylinux-9 https://zuul.opendev.org/t/openstack/build/8c8b5f6b5f1342d2bb9a9501308cb996 : FAILURE in 1h 13m 27s\n- openstack-ansible-upgrade-infra_lxc-ubuntu-jammy https://zuul.opendev.org/t/openstack/build/50b6e8b671cb43d881685f917c95af2c : FAILURE in 1h 09m 10s\n- openstack-ansible-deploy-hosts_lxc_zfs-ubuntu-jammy https://zuul.opendev.org/t/openstack/build/b02684fcd3104092a154b142e18f15e8 : FAILURE in 19m 53s\n- openstack-ansible-deploy-hosts_lxc_btrfs-ubuntu-jammy https://zuul.opendev.org/t/openstack/build/5e52b2e2d62346199434f47f1df671ca : FAILURE in 20m 25s\n- openstack-ansible-deploy-hosts_lxc_overlayfs-ubuntu-jammy https://zuul.opendev.org/t/openstack/build/94f2ec9074da45e78b4267ab7dff3b12 : FAILURE in 18m 37s\n- openstack-ansible-deploy-hosts_lxc_btrfs-debian-bullseye https://zuul.opendev.org/t/openstack/build/a6b91695e195456b8f3480dbab7eb343 : FAILURE in 20m 00s\n- openstack-ansible-deploy-hosts_lxc_overlayfs-debian-bullseye https://zuul.opendev.org/t/openstack/build/1411aa4e6738483f8033084e41c376e2 : FAILURE in 25m 09s\n- openstack-ansible-deploy-hosts_lxc_overlayfs-rockylinux-9 https://zuul.opendev.org/t/openstack/build/1e63b74774924e4ea77c91012f33b3ea : FAILURE in 28m 41s","accounts_in_message":[],"_revision_number":2},{"id":"7e8d7ea266b2d58c935bc1f941fe55c659752e4c","tag":"autogenerated:gerrit:newPatchSet","author":{"_account_id":25023,"name":"Jonathan Rosser","email":"jonathan.rosser@rd.bbc.co.uk","username":"jrosser"},"date":"2023-10-11 15:56:08.000000000","message":"Patch Set 3: Published edit on patch set 2.\n\nOutdated Votes:\n* Verified-1\n","accounts_in_message":[],"_revision_number":3},{"id":"5b860478827f4b8426d2080841b990357ad959b4","tag":"autogenerated:zuul:check","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2023-10-11 18:13:05.000000000","message":"Patch Set 3: Verified+1\n\nBuild succeeded (check pipeline).\nhttps://zuul.opendev.org/t/openstack/buildset/b0c582a96b5e4f16bff8531379830191\n\n- openstack-ansible-linters-ubuntu-jammy https://zuul.opendev.org/t/openstack/build/334f722507a1447e89305151a296a0f1 : SUCCESS in 13m 54s\n- openstack-tox-docs https://zuul.opendev.org/t/openstack/build/7c9e53c275694d71980e2d7486047572 : SUCCESS in 5m 16s\n- build-openstack-releasenotes https://zuul.opendev.org/t/openstack/build/8a933cd3c3094b6bafadfae3cb94d797 : SUCCESS in 2m 55s\n- openstack-ansible-deploy-infra_lxc-centos-9-stream https://zuul.opendev.org/t/openstack/build/b4d944fc5b2642cf8a255782e4ebd000 : SUCCESS in 1h 38m 50s (non-voting)\n- openstack-ansible-deploy-infra_lxc-rockylinux-9 https://zuul.opendev.org/t/openstack/build/ceac394a3e8944a28f9ef330d7acbd36 : SUCCESS in 1h 38m 29s\n- openstack-ansible-deploy-infra_lxc_tls-rockylinux-9 https://zuul.opendev.org/t/openstack/build/e4f0a7b41eb44b568ed0a0109af191c7 : SUCCESS in 1h 39m 33s\n- openstack-ansible-deploy-infra_lxc-debian-bullseye https://zuul.opendev.org/t/openstack/build/998ae69d6d69484db4c0831922a7bfb4 : SUCCESS in 1h 24m 41s\n- openstack-ansible-deploy-infra_lxc-ubuntu-jammy https://zuul.opendev.org/t/openstack/build/3121a1c5d619439f98fdd53ded08dd11 : SUCCESS in 1h 28m 18s\n- openstack-ansible-deploy-infra_lxc_stepca-ubuntu-jammy https://zuul.opendev.org/t/openstack/build/b2afa87c77ca482f9d2ab87bae958677 : SUCCESS in 1h 31m 23s\n- openstack-ansible-deploy-infra_lxc_tls-ubuntu-jammy https://zuul.opendev.org/t/openstack/build/9a1b24346d80437bbded0fe9927374f4 : SUCCESS in 1h 21m 32s\n- openstack-ansible-upgrade-infra_lxc-rockylinux-9 https://zuul.opendev.org/t/openstack/build/f80f26a4b2304db8a8f35c329f12e9cf : SUCCESS in 2h 06m 32s\n- openstack-ansible-upgrade-infra_lxc-ubuntu-jammy https://zuul.opendev.org/t/openstack/build/0abbaa4e5f074049a42356a4931766c5 : SUCCESS in 1h 50m 08s\n- openstack-ansible-deploy-hosts_lxc_zfs-ubuntu-jammy https://zuul.opendev.org/t/openstack/build/bf0a6f79c78741548aaf5dcfc03adff6 : SUCCESS in 33m 04s\n- openstack-ansible-deploy-hosts_lxc_btrfs-ubuntu-jammy https://zuul.opendev.org/t/openstack/build/4b08e74b0eed4fa28a1360f7f1638b45 : SUCCESS in 33m 30s\n- openstack-ansible-deploy-hosts_lxc_overlayfs-ubuntu-jammy https://zuul.opendev.org/t/openstack/build/5922a6717c8e4fa38fa332b6d0a169d0 : SUCCESS in 29m 33s\n- openstack-ansible-deploy-hosts_lxc_btrfs-debian-bullseye https://zuul.opendev.org/t/openstack/build/70017976ec0046edb3fb16ec6ad19e50 : SUCCESS in 29m 40s\n- openstack-ansible-deploy-hosts_lxc_overlayfs-debian-bullseye https://zuul.opendev.org/t/openstack/build/f5043bfc565845ddbda0e2401083a9d3 : SUCCESS in 32m 38s\n- openstack-ansible-deploy-hosts_lxc_overlayfs-rockylinux-9 https://zuul.opendev.org/t/openstack/build/2756b2c1144a415e83fe90975581cdd7 : SUCCESS in 31m 54s","accounts_in_message":[],"_revision_number":3},{"id":"e07413b4904b5bbea7475fdcbe412a3059da9667","author":{"_account_id":28619,"name":"Dmitriy Rabotyagov","email":"noonedeadpunk@gmail.com","username":"noonedeadpunk"},"date":"2023-10-11 18:17:04.000000000","message":"Patch Set 3: Code-Review+2","accounts_in_message":[],"_revision_number":3},{"id":"206b47095f90e80e05dba193554cfc2159662c74","author":{"_account_id":34411,"name":"Neil Hanlon","email":"neil@shrug.pw","username":"nhanlon"},"date":"2023-10-12 02:58:34.000000000","message":"Patch Set 3: Code-Review+2 Workflow+1","accounts_in_message":[],"_revision_number":3},{"id":"e3064ea9a9a76424180a8dc58ddd76e7503bf0de","tag":"autogenerated:zuul:gate","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2023-10-12 03:01:17.000000000","message":"Patch Set 3: -Verified\n\nStarting gate jobs.","accounts_in_message":[],"_revision_number":3},{"id":"20f7faf012ca3adfc261687d4d022eb3d0691bc7","tag":"autogenerated:zuul:gate","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2023-10-12 05:03:32.000000000","message":"Patch Set 3: Verified+2\n\nBuild succeeded (gate pipeline).\nhttps://zuul.opendev.org/t/openstack/buildset/c4d66d96e34b462c9a18dab3ffe3d051\n\n- openstack-ansible-linters-ubuntu-jammy https://zuul.opendev.org/t/openstack/build/ec0d7c78128d44be88ff4f66e6abe225 : SUCCESS in 13m 49s\n- openstack-tox-docs https://zuul.opendev.org/t/openstack/build/fcbca4b0861643b3a4b589ee7d1e7b5d : SUCCESS in 5m 13s\n- build-openstack-releasenotes https://zuul.opendev.org/t/openstack/build/891ca5dd0413442c8575305e272fcb07 : SUCCESS in 3m 28s\n- openstack-ansible-deploy-infra_lxc-rockylinux-9 https://zuul.opendev.org/t/openstack/build/18416692e12e4aeca656282aad8eb902 : SUCCESS in 1h 41m 26s\n- openstack-ansible-deploy-infra_lxc_tls-rockylinux-9 https://zuul.opendev.org/t/openstack/build/c27a828100c54742adfc4884201acdf1 : SUCCESS in 1h 38m 00s\n- openstack-ansible-deploy-infra_lxc-debian-bullseye https://zuul.opendev.org/t/openstack/build/a6fea32b6181407ab2096e0d12dc3558 : SUCCESS in 1h 05m 39s\n- openstack-ansible-deploy-infra_lxc-ubuntu-jammy https://zuul.opendev.org/t/openstack/build/20519789316b4ff0a1504c3c7fad68ad : SUCCESS in 1h 10m 52s\n- openstack-ansible-deploy-infra_lxc_stepca-ubuntu-jammy https://zuul.opendev.org/t/openstack/build/59a01f3f352e4eeabedb72121de2e8df : SUCCESS in 1h 24m 59s\n- openstack-ansible-deploy-infra_lxc_tls-ubuntu-jammy https://zuul.opendev.org/t/openstack/build/cfbc89b207f8434fab4885369face4b2 : SUCCESS in 1h 20m 48s\n- openstack-ansible-upgrade-infra_lxc-rockylinux-9 https://zuul.opendev.org/t/openstack/build/eb613d48d1714561b6cb1ad53aa672a5 : SUCCESS in 1h 58m 24s\n- openstack-ansible-upgrade-infra_lxc-ubuntu-jammy https://zuul.opendev.org/t/openstack/build/50ba4344a8554372a243320be277d6de : SUCCESS in 1h 37m 42s\n- openstack-ansible-deploy-hosts_lxc_zfs-ubuntu-jammy https://zuul.opendev.org/t/openstack/build/2493a2e84a5945a788d0c831c76872ad : SUCCESS in 25m 18s\n- openstack-ansible-deploy-hosts_lxc_btrfs-ubuntu-jammy https://zuul.opendev.org/t/openstack/build/789bf85ee0d3447eb71fd211e5a0de6f : SUCCESS in 26m 01s\n- openstack-ansible-deploy-hosts_lxc_overlayfs-ubuntu-jammy https://zuul.opendev.org/t/openstack/build/dfb4501ebe184adf8a3d8ddb6361c839 : SUCCESS in 25m 16s\n- openstack-ansible-deploy-hosts_lxc_btrfs-debian-bullseye https://zuul.opendev.org/t/openstack/build/d2bd90a9df2544938bf0712a99c48b37 : SUCCESS in 27m 59s\n- openstack-ansible-deploy-hosts_lxc_overlayfs-debian-bullseye https://zuul.opendev.org/t/openstack/build/c3fc192ae78e43deaacbd706197115b5 : SUCCESS in 24m 17s\n- openstack-ansible-deploy-hosts_lxc_overlayfs-rockylinux-9 https://zuul.opendev.org/t/openstack/build/44fc4b2f9115406295af9e8b80f83fcb : SUCCESS in 30m 50s","accounts_in_message":[],"_revision_number":3},{"id":"3074dcd9b945026ed7cbb92ab53973a7a301140e","tag":"autogenerated:gerrit:merged","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2023-10-12 05:03:32.000000000","message":"Change has been successfully merged","accounts_in_message":[],"_revision_number":3},{"id":"893c89197f0a342aeb3c13c1c392cec61c598173","tag":"autogenerated:zuul:promote","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2023-10-12 05:05:16.000000000","message":"Patch Set 3:\n\nBuild succeeded (promote pipeline).\nhttps://zuul.opendev.org/t/openstack/buildset/655b658b4cb347849f42534d5e7f58ec\n\n- promote-openstack-tox-docs https://zuul.opendev.org/t/openstack/build/92a6a9d7bd604d879858153fe218ccdc : SUCCESS in 1m 26s","accounts_in_message":[],"_revision_number":3}],"current_revision_number":3,"current_revision":"d57f9a8f47a4821c9d843c3007a931315a40e3fc","revisions":{"00ed745f573553ec553c90ed78ac1638e3895ffb":{"kind":"REWORK","_number":1,"created":"2023-10-10 16:51:42.000000000","uploader":{"_account_id":25023,"name":"Jonathan Rosser","email":"jonathan.rosser@rd.bbc.co.uk","username":"jrosser"},"ref":"refs/changes/61/897861/1","fetch":{"anonymous http":{"url":"https://review.opendev.org/openstack/openstack-ansible-lxc_hosts","ref":"refs/changes/61/897861/1","commands":{"Checkout":"git fetch https://review.opendev.org/openstack/openstack-ansible-lxc_hosts refs/changes/61/897861/1 \u0026\u0026 git checkout FETCH_HEAD","Cherry Pick":"git fetch https://review.opendev.org/openstack/openstack-ansible-lxc_hosts refs/changes/61/897861/1 \u0026\u0026 git cherry-pick FETCH_HEAD","Format Patch":"git fetch https://review.opendev.org/openstack/openstack-ansible-lxc_hosts refs/changes/61/897861/1 \u0026\u0026 git format-patch -1 --stdout FETCH_HEAD","Pull":"git pull https://review.opendev.org/openstack/openstack-ansible-lxc_hosts refs/changes/61/897861/1"}}},"commit":{"parents":[{"commit":"a22ec2150f6b9300cbe0dd9f1acb84f034056bf4","subject":"Remove old tasks and vars from image download process","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/openstack-ansible-lxc_hosts/commit/a22ec2150f6b9300cbe0dd9f1acb84f034056bf4"}]}],"author":{"name":"Jonathan Rosser","email":"jonathan.rosser@rd.bbc.co.uk","date":"2023-10-10 16:50:54.000000000","tz":60},"committer":{"name":"Jonathan Rosser","email":"jonathan.rosser@rd.bbc.co.uk","date":"2023-10-10 16:50:54.000000000","tz":60},"subject":"Remove lxc_cache_map variable","message":"Remove lxc_cache_map variable\n\nThis has not had any practical use for several releases and mostly\ncarries copies of ansible facts. Remove the variable and use the\nfacts directly.\n\nChange-Id: I1d2be9d07b38eaf2b737819c451a0d2339f723d0\n","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/openstack-ansible-lxc_hosts/commit/00ed745f573553ec553c90ed78ac1638e3895ffb"}],"resolve_conflicts_web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/openstack-ansible-lxc_hosts/commit/00ed745f573553ec553c90ed78ac1638e3895ffb"}]},"branch":"refs/heads/master"},"d14816157ce5573f8d17e18ca85853edd3710a41":{"kind":"REWORK","_number":2,"created":"2023-10-11 13:22:26.000000000","uploader":{"_account_id":25023,"name":"Jonathan Rosser","email":"jonathan.rosser@rd.bbc.co.uk","username":"jrosser"},"ref":"refs/changes/61/897861/2","fetch":{"anonymous http":{"url":"https://review.opendev.org/openstack/openstack-ansible-lxc_hosts","ref":"refs/changes/61/897861/2","commands":{"Checkout":"git fetch https://review.opendev.org/openstack/openstack-ansible-lxc_hosts refs/changes/61/897861/2 \u0026\u0026 git checkout FETCH_HEAD","Cherry Pick":"git fetch https://review.opendev.org/openstack/openstack-ansible-lxc_hosts refs/changes/61/897861/2 \u0026\u0026 git cherry-pick FETCH_HEAD","Format Patch":"git fetch https://review.opendev.org/openstack/openstack-ansible-lxc_hosts refs/changes/61/897861/2 \u0026\u0026 git format-patch -1 --stdout FETCH_HEAD","Pull":"git pull https://review.opendev.org/openstack/openstack-ansible-lxc_hosts refs/changes/61/897861/2"}}},"commit":{"parents":[{"commit":"a22ec2150f6b9300cbe0dd9f1acb84f034056bf4","subject":"Remove old tasks and vars from image download process","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/openstack-ansible-lxc_hosts/commit/a22ec2150f6b9300cbe0dd9f1acb84f034056bf4"}]}],"author":{"name":"Jonathan Rosser","email":"jonathan.rosser@rd.bbc.co.uk","date":"2023-10-10 16:50:54.000000000","tz":60},"committer":{"name":"Jonathan Rosser","email":"jonathan.rosser@rd.bbc.co.uk","date":"2023-10-11 13:22:26.000000000","tz":0},"subject":"Remove lxc_cache_map variable","message":"Remove lxc_cache_map variable\n\nThis has not had any practical use for several releases and mostly\ncarries copies of ansible facts. Remove the variable and use the\nfacts directly.\n\nChange-Id: I1d2be9d07b38eaf2b737819c451a0d2339f723d0\n","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/openstack-ansible-lxc_hosts/commit/d14816157ce5573f8d17e18ca85853edd3710a41"}],"resolve_conflicts_web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/openstack-ansible-lxc_hosts/commit/d14816157ce5573f8d17e18ca85853edd3710a41"}]},"branch":"refs/heads/master"},"d57f9a8f47a4821c9d843c3007a931315a40e3fc":{"kind":"REWORK","_number":3,"created":"2023-10-11 15:56:08.000000000","uploader":{"_account_id":25023,"name":"Jonathan Rosser","email":"jonathan.rosser@rd.bbc.co.uk","username":"jrosser"},"ref":"refs/changes/61/897861/3","fetch":{"anonymous http":{"url":"https://review.opendev.org/openstack/openstack-ansible-lxc_hosts","ref":"refs/changes/61/897861/3","commands":{"Checkout":"git fetch https://review.opendev.org/openstack/openstack-ansible-lxc_hosts refs/changes/61/897861/3 \u0026\u0026 git checkout FETCH_HEAD","Cherry Pick":"git fetch https://review.opendev.org/openstack/openstack-ansible-lxc_hosts refs/changes/61/897861/3 \u0026\u0026 git cherry-pick FETCH_HEAD","Format Patch":"git fetch https://review.opendev.org/openstack/openstack-ansible-lxc_hosts refs/changes/61/897861/3 \u0026\u0026 git format-patch -1 --stdout FETCH_HEAD","Pull":"git pull https://review.opendev.org/openstack/openstack-ansible-lxc_hosts refs/changes/61/897861/3"}}},"commit":{"parents":[{"commit":"a22ec2150f6b9300cbe0dd9f1acb84f034056bf4","subject":"Remove old tasks and vars from image download process","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/openstack-ansible-lxc_hosts/commit/a22ec2150f6b9300cbe0dd9f1acb84f034056bf4"}]}],"author":{"name":"Jonathan Rosser","email":"jonathan.rosser@rd.bbc.co.uk","date":"2023-10-10 16:50:54.000000000","tz":60},"committer":{"name":"Jonathan Rosser","email":"jonathan.rosser@rd.bbc.co.uk","date":"2023-10-11 15:56:08.000000000","tz":0},"subject":"Remove lxc_cache_map variable","message":"Remove lxc_cache_map variable\n\nThis has not had any practical use for several releases and mostly\ncarries copies of ansible facts. Remove the variable and use the\nfacts directly.\n\nChange-Id: I1d2be9d07b38eaf2b737819c451a0d2339f723d0\n","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/openstack-ansible-lxc_hosts/commit/d57f9a8f47a4821c9d843c3007a931315a40e3fc"}],"resolve_conflicts_web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/openstack-ansible-lxc_hosts/commit/d57f9a8f47a4821c9d843c3007a931315a40e3fc"}]},"branch":"refs/heads/master"}},"requirements":[],"submit_records":[{"rule_name":"gerrit~DefaultSubmitRule","status":"CLOSED","labels":[{"label":"Verified","status":"MAY","applied_by":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]}},{"label":"Code-Review","status":"MAY","applied_by":{"_account_id":28619,"name":"Dmitriy Rabotyagov","email":"noonedeadpunk@gmail.com","username":"noonedeadpunk"}},{"label":"Workflow","status":"MAY","applied_by":{"_account_id":34411,"name":"Neil Hanlon","email":"neil@shrug.pw","username":"nhanlon"}},{"label":"Backport-Candidate","status":"MAY"}]}],"submit_requirements":[{"name":"Verified","description":"Verified in gate by CI","status":"SATISFIED","is_legacy":false,"submittability_expression_result":{"expression":"label:Verified\u003dMAX AND -label:Verified\u003dMIN","fulfilled":true,"status":"PASS","passing_atoms":["label:Verified\u003dMAX"],"failing_atoms":["label:Verified\u003dMIN"],"atom_explanations":{}}},{"name":"Backport-Candidate","description":"Backport candidate status","status":"NOT_APPLICABLE","is_legacy":false,"applicability_expression_result":{"fulfilled":false,"status":"FAIL"},"submittability_expression_result":{"expression":"is:true","fulfilled":true,"status":"NOT_EVALUATED","passing_atoms":[],"failing_atoms":[],"atom_explanations":{}}},{"name":"Code-Review","description":"Code reviewed by core reviewer","status":"SATISFIED","is_legacy":false,"submittability_expression_result":{"expression":"label:Code-Review\u003dMAX AND -label:Code-Review\u003dMIN","fulfilled":true,"status":"PASS","passing_atoms":["label:Code-Review\u003dMAX"],"failing_atoms":["label:Code-Review\u003dMIN"],"atom_explanations":{}}},{"name":"Workflow","description":"Approved for gate by core reviewer","status":"SATISFIED","is_legacy":false,"submittability_expression_result":{"expression":"label:Workflow\u003dMAX AND -label:Workflow\u003dMIN","fulfilled":true,"status":"PASS","passing_atoms":["label:Workflow\u003dMAX"],"failing_atoms":["label:Workflow\u003dMIN"],"atom_explanations":{}}}]}
