)]}'
{"id":"openstack%2Fopenstack-ansible-lxc_hosts~924340","triplet_id":"openstack%2Fopenstack-ansible-lxc_hosts~master~I7fe143f1812cd2fe3b50c9a327cec44f4523c1ab","project":"openstack/openstack-ansible-lxc_hosts","branch":"master","topic":"osa/noble","attention_set":{},"removed_from_attention_set":{"25023":{"account":{"_account_id":25023,"name":"Jonathan Rosser","email":"jonathan.rosser@rd.bbc.co.uk","username":"jrosser"},"last_update":"2024-07-24 13:56:12.000000000","reason":"Change was submitted"}},"hashtags":[],"change_id":"I7fe143f1812cd2fe3b50c9a327cec44f4523c1ab","subject":"Use generated apparmor profile by default in lxc base config","status":"MERGED","created":"2024-07-17 15:26:05.000000000","updated":"2024-07-24 13:57:20.000000000","submitted":"2024-07-24 13:56:12.000000000","submitter":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"total_comment_count":2,"unresolved_comment_count":0,"has_review_started":true,"submission_id":"924340-osa/noble","meta_rev_id":"480159fad1b36b7f4eccb58294985d45d465d823","_number":924340,"virtual_id_number":924340,"owner":{"_account_id":25023,"name":"Jonathan Rosser","email":"jonathan.rosser@rd.bbc.co.uk","username":"jrosser"},"actions":{},"labels":{"Verified":{"approved":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"all":[{"value":0,"_account_id":31542,"name":"Andrew Bonney","email":"andrew.bonney@bbc.co.uk","username":"andrewbonney"},{"value":0,"_account_id":28619,"name":"Dmitriy Rabotyagov","email":"noonedeadpunk@gmail.com","username":"noonedeadpunk"},{"tag":"autogenerated:zuul:gate","value":2,"date":"2024-07-24 13:56:12.000000000","permitted_voting_range":{"min":2,"max":2},"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]}],"values":{"-2":"Fails","-1":"Doesn\u0027t seem to work"," 0":"No score","+1":"Works for me","+2":"Verified"},"description":"","default_value":0,"optional":true},"Code-Review":{"approved":{"_account_id":31542,"name":"Andrew Bonney","email":"andrew.bonney@bbc.co.uk","username":"andrewbonney"},"all":[{"value":2,"date":"2024-07-23 12:50:29.000000000","permitted_voting_range":{"min":2,"max":2},"_account_id":31542,"name":"Andrew Bonney","email":"andrew.bonney@bbc.co.uk","username":"andrewbonney"},{"value":2,"date":"2024-07-22 11:08:48.000000000","permitted_voting_range":{"min":2,"max":2},"_account_id":28619,"name":"Dmitriy Rabotyagov","email":"noonedeadpunk@gmail.com","username":"noonedeadpunk"},{"value":0,"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]}],"values":{"-2":"Do not merge","-1":"This patch needs further work before it can be merged"," 0":"No score","+1":"Looks good to me, but someone else must approve","+2":"Looks good to me (core reviewer)"},"description":"","default_value":0,"optional":true},"Workflow":{"approved":{"_account_id":28619,"name":"Dmitriy Rabotyagov","email":"noonedeadpunk@gmail.com","username":"noonedeadpunk"},"all":[{"value":0,"_account_id":31542,"name":"Andrew Bonney","email":"andrew.bonney@bbc.co.uk","username":"andrewbonney"},{"value":1,"date":"2024-07-24 11:59:54.000000000","permitted_voting_range":{"min":1,"max":1},"_account_id":28619,"name":"Dmitriy Rabotyagov","email":"noonedeadpunk@gmail.com","username":"noonedeadpunk"},{"value":0,"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]}],"values":{"-1":"Work in progress"," 0":"Ready for reviews","+1":"Approved"},"description":"","default_value":0,"optional":true},"Backport-Candidate":{"all":[{"value":0,"_account_id":31542,"name":"Andrew Bonney","email":"andrew.bonney@bbc.co.uk","username":"andrewbonney"},{"value":0,"_account_id":28619,"name":"Dmitriy Rabotyagov","email":"noonedeadpunk@gmail.com","username":"noonedeadpunk"},{"value":0,"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]}],"values":{"-1":"Not A Backport Candidate"," 0":"Backport Review Needed","+1":"Proposed Backport"},"description":"","default_value":0,"optional":true}},"removable_reviewers":[],"reviewers":{"REVIEWER":[{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},{"_account_id":28619,"name":"Dmitriy Rabotyagov","email":"noonedeadpunk@gmail.com","username":"noonedeadpunk"},{"_account_id":31542,"name":"Andrew Bonney","email":"andrew.bonney@bbc.co.uk","username":"andrewbonney"}]},"pending_reviewers":{},"reviewer_updates":[{"updated":"2024-07-17 17:36:25.000000000","updated_by":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"reviewer":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"state":"REVIEWER"},{"updated":"2024-07-22 11:08:48.000000000","updated_by":{"_account_id":28619,"name":"Dmitriy Rabotyagov","email":"noonedeadpunk@gmail.com","username":"noonedeadpunk"},"reviewer":{"_account_id":28619,"name":"Dmitriy Rabotyagov","email":"noonedeadpunk@gmail.com","username":"noonedeadpunk"},"state":"REVIEWER"},{"updated":"2024-07-23 12:50:29.000000000","updated_by":{"_account_id":31542,"name":"Andrew Bonney","email":"andrew.bonney@bbc.co.uk","username":"andrewbonney"},"reviewer":{"_account_id":31542,"name":"Andrew Bonney","email":"andrew.bonney@bbc.co.uk","username":"andrewbonney"},"state":"REVIEWER"}],"messages":[{"id":"81bac971a20d74052515303a8f0f254a7e0a4daa","tag":"autogenerated:gerrit:newPatchSet","author":{"_account_id":25023,"name":"Jonathan Rosser","email":"jonathan.rosser@rd.bbc.co.uk","username":"jrosser"},"date":"2024-07-17 15:26:05.000000000","message":"Uploaded patch set 1.","accounts_in_message":[],"_revision_number":1},{"id":"8997c12b41201ceaa6418385e1f6727bf5b1b560","tag":"autogenerated:gerrit:setTopic","author":{"_account_id":25023,"name":"Jonathan Rosser","email":"jonathan.rosser@rd.bbc.co.uk","username":"jrosser"},"date":"2024-07-17 15:46:20.000000000","message":"Topic set to osa/noble","accounts_in_message":[],"_revision_number":1},{"id":"f5d1cc8376517d07b9360d13f97150377822659c","tag":"autogenerated:zuul:check","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2024-07-17 17:36:25.000000000","message":"Patch Set 1: Verified-1\n\nBuild failed (check pipeline).  For information on how to proceed, see\nhttps://docs.opendev.org/opendev/infra-manual/latest/developers.html#automated-testing\nand https://docs.openstack.org/project-team-guide/testing.html#how-to-handle-test-failures\n\nhttps://zuul.opendev.org/t/openstack/buildset/86aed672f61b43f5bc507196d5b40fbf\n\n- openstack-ansible-linters-ubuntu-jammy https://zuul.opendev.org/t/openstack/build/479b223f377140d6a9f8b9038409d28f : SUCCESS in 16m 57s\n- openstack-tox-docs https://zuul.opendev.org/t/openstack/build/4406690100a5452eb08b83fbdeb0021a : SUCCESS in 6m 28s\n- openstack-ansible-deploy-infra_lxc-centos-9-stream https://zuul.opendev.org/t/openstack/build/27a241ec58214dcbae62070e065a1a7e : FAILURE in 1h 13m 08s (non-voting)\n- openstack-ansible-deploy-infra_lxc-rockylinux-9 https://zuul.opendev.org/t/openstack/build/21abce69f1f04aff94dd3b4d4f3f1177 : FAILURE in 1h 09m 50s\n- openstack-ansible-deploy-infra_lxc_tls-rockylinux-9 https://zuul.opendev.org/t/openstack/build/48ca7ce6f8e9430780c468a954534447 : FAILURE in 1h 11m 23s\n- openstack-ansible-deploy-infra_lxc-debian-bookworm https://zuul.opendev.org/t/openstack/build/a244f66fa73a4b3d917e3e65463f8b28 : SUCCESS in 1h 13m 57s\n- openstack-ansible-deploy-infra_lxc-ubuntu-jammy https://zuul.opendev.org/t/openstack/build/eab7af76084b45268ab95af2d5d996f5 : FAILURE in 54m 50s\n- openstack-ansible-deploy-infra_lxc_stepca-ubuntu-jammy https://zuul.opendev.org/t/openstack/build/98e7970cde6b4133b89bc90ca58a293d : FAILURE in 57m 02s\n- openstack-ansible-deploy-infra_lxc_tls-ubuntu-jammy https://zuul.opendev.org/t/openstack/build/f773d0075dbb473b81a37b399890e133 : FAILURE in 57m 58s\n- openstack-ansible-upgrade-infra_lxc-rockylinux-9 https://zuul.opendev.org/t/openstack/build/d34c5eaa671a40dbb9ca12f9455b992f : SUCCESS in 2h 03m 22s\n- openstack-ansible-upgrade-infra_lxc-ubuntu-jammy https://zuul.opendev.org/t/openstack/build/0037e794d10d443a859d7802e7584303 : SUCCESS in 1h 33m 48s\n- openstack-ansible-deploy-hosts_lxc_zfs-ubuntu-jammy https://zuul.opendev.org/t/openstack/build/e1484b2f515f482ca9ab2a2065abcba9 : SUCCESS in 24m 15s\n- openstack-ansible-deploy-hosts_lxc_btrfs-ubuntu-jammy https://zuul.opendev.org/t/openstack/build/876965355d244c0f9193e874c5548d1d : SUCCESS in 23m 23s\n- openstack-ansible-deploy-hosts_lxc_overlayfs-ubuntu-jammy https://zuul.opendev.org/t/openstack/build/a56e6b4c802a4e9aa4337ea842a8d60e : SUCCESS in 25m 06s\n- openstack-ansible-deploy-hosts_lxc_btrfs-debian-bookworm https://zuul.opendev.org/t/openstack/build/08c48c4ba1c14b179c7e1ff2363c8a1d : SUCCESS in 25m 29s\n- openstack-ansible-deploy-hosts_lxc_overlayfs-debian-bookworm https://zuul.opendev.org/t/openstack/build/18e3c6e021bb4bcb8cc1c912e3b6bc4a : SUCCESS in 26m 22s\n- openstack-ansible-deploy-hosts_lxc_overlayfs-rockylinux-9 https://zuul.opendev.org/t/openstack/build/144650b73c294abe811179bd6dcecbf6 : SUCCESS in 28m 13s","accounts_in_message":[],"_revision_number":1},{"id":"157022b0f374fad6d5ba8f901d9949e75f300079","author":{"_account_id":25023,"name":"Jonathan Rosser","email":"jonathan.rosser@rd.bbc.co.uk","username":"jrosser"},"date":"2024-07-17 20:26:15.000000000","message":"Patch Set 1:\n\n(1 comment)","accounts_in_message":[],"_revision_number":1},{"id":"4c7f7ad87ac83e34c9a6c9aeb35540e228c42182","tag":"autogenerated:zuul:check","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2024-07-17 23:01:44.000000000","message":"Patch Set 1:\n\nBuild failed (check pipeline).  For information on how to proceed, see\nhttps://docs.opendev.org/opendev/infra-manual/latest/developers.html#automated-testing\nand https://docs.openstack.org/project-team-guide/testing.html#how-to-handle-test-failures\n\nhttps://zuul.opendev.org/t/openstack/buildset/ce2c9b714468474b9ecdf3ca220f8ab4\n\n- openstack-ansible-linters-ubuntu-jammy https://zuul.opendev.org/t/openstack/build/f7dcb9963eec4b31a421b5c29b5b5660 : SUCCESS in 13m 18s\n- openstack-tox-docs https://zuul.opendev.org/t/openstack/build/4b53e4df83224608aaef3e0e37fd162d : SUCCESS in 5m 38s\n- openstack-ansible-deploy-infra_lxc-centos-9-stream https://zuul.opendev.org/t/openstack/build/d5a3472cf92f408884628f96d5c37950 : SUCCESS in 1h 17m 57s (non-voting)\n- openstack-ansible-deploy-infra_lxc-rockylinux-9 https://zuul.opendev.org/t/openstack/build/fce17979435a4d3080a453010df5568b : SUCCESS in 1h 22m 07s\n- openstack-ansible-deploy-infra_lxc_tls-rockylinux-9 https://zuul.opendev.org/t/openstack/build/0e45df16798f4b96b1afba07e71093c8 : FAILURE in 1h 11m 58s\n- openstack-ansible-deploy-infra_lxc-debian-bookworm https://zuul.opendev.org/t/openstack/build/d1aeeab0cb6545e2967a129b0d187505 : FAILURE in 1h 07m 33s\n- openstack-ansible-deploy-infra_lxc-ubuntu-jammy https://zuul.opendev.org/t/openstack/build/b785e0001bc34ce7bd156dec5b4795b1 : SUCCESS in 1h 09m 15s\n- openstack-ansible-deploy-infra_lxc_stepca-ubuntu-jammy https://zuul.opendev.org/t/openstack/build/6fe3e19f500649cb83b73ba40b997abd : FAILURE in 1h 00m 18s\n- openstack-ansible-deploy-infra_lxc_tls-ubuntu-jammy https://zuul.opendev.org/t/openstack/build/2786ec11f2404fbfb75257fb221051a9 : FAILURE in 1h 08m 21s\n- openstack-ansible-upgrade-infra_lxc-rockylinux-9 https://zuul.opendev.org/t/openstack/build/0b0ae079991d48d6a294f35ab533c88c : SUCCESS in 1h 48m 09s\n- openstack-ansible-upgrade-infra_lxc-ubuntu-jammy https://zuul.opendev.org/t/openstack/build/28b351d10472491488d52a3337b65df8 : SUCCESS in 1h 20m 26s\n- openstack-ansible-deploy-hosts_lxc_zfs-ubuntu-jammy https://zuul.opendev.org/t/openstack/build/9b54cc5bcc80450b9a95c2981a54ff64 : SUCCESS in 28m 59s\n- openstack-ansible-deploy-hosts_lxc_btrfs-ubuntu-jammy https://zuul.opendev.org/t/openstack/build/48b01dbbce0c447893d0d1f5affa0e73 : SUCCESS in 31m 50s\n- openstack-ansible-deploy-hosts_lxc_overlayfs-ubuntu-jammy https://zuul.opendev.org/t/openstack/build/124de86f55cf43eca574844fc2fca9ea : SUCCESS in 23m 22s\n- openstack-ansible-deploy-hosts_lxc_btrfs-debian-bookworm https://zuul.opendev.org/t/openstack/build/c129c5b329df4b9585519ac2f9793eae : SUCCESS in 24m 50s\n- openstack-ansible-deploy-hosts_lxc_overlayfs-debian-bookworm https://zuul.opendev.org/t/openstack/build/ccfa87d2fb74473680e552283beb4869 : SUCCESS in 23m 40s\n- openstack-ansible-deploy-hosts_lxc_overlayfs-rockylinux-9 https://zuul.opendev.org/t/openstack/build/5ae4f50ac58746f69cbed4a84a532d30 : SUCCESS in 28m 25s","accounts_in_message":[],"_revision_number":1},{"id":"df8cc345c354feb79bdac657deeb336cd6009e18","tag":"autogenerated:gerrit:newPatchSet","author":{"_account_id":25023,"name":"Jonathan Rosser","email":"jonathan.rosser@rd.bbc.co.uk","username":"jrosser"},"date":"2024-07-19 19:10:32.000000000","message":"Patch Set 2: Patch Set 1 was rebased\n\nOutdated Votes:\n* Verified-1\n","accounts_in_message":[],"_revision_number":2},{"id":"4b5e74f3b77f9a06f7ab37324bffc0ca0e09fa17","tag":"autogenerated:zuul:check","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2024-07-19 21:17:53.000000000","message":"Patch Set 2: Verified-1\n\nBuild failed (check pipeline).  For information on how to proceed, see\nhttps://docs.opendev.org/opendev/infra-manual/latest/developers.html#automated-testing\nand https://docs.openstack.org/project-team-guide/testing.html#how-to-handle-test-failures\n\nhttps://zuul.opendev.org/t/openstack/buildset/cd68a1d71c05459986dc784590ceb495\n\n- openstack-ansible-linters-ubuntu-jammy https://zuul.opendev.org/t/openstack/build/14b72b5366554e5aa9eaebdd82773280 : SUCCESS in 11m 52s\n- openstack-tox-docs https://zuul.opendev.org/t/openstack/build/ebe67a18e78b4cba8ab84da7b8c371aa : SUCCESS in 4m 12s\n- openstack-ansible-deploy-infra_lxc-centos-9-stream https://zuul.opendev.org/t/openstack/build/f3649a96c5b44fe6a550ed54bfdb4d8f : SUCCESS in 1h 26m 05s (non-voting)\n- openstack-ansible-deploy-infra_lxc-rockylinux-9 https://zuul.opendev.org/t/openstack/build/a3548d7848844e4fb1eb74492f45c1ef : SUCCESS in 1h 26m 56s\n- openstack-ansible-deploy-infra_lxc_tls-rockylinux-9 https://zuul.opendev.org/t/openstack/build/b929c69286984c0e92e9b3977e29b78c : SUCCESS in 1h 25m 17s\n- openstack-ansible-deploy-infra_lxc-debian-bookworm https://zuul.opendev.org/t/openstack/build/362f36f2e1264f3abf97dab590d49a32 : SUCCESS in 1h 21m 03s\n- openstack-ansible-deploy-infra_lxc-ubuntu-jammy https://zuul.opendev.org/t/openstack/build/7dc952cb256f45a3b1c406eea9e8f1c1 : SUCCESS in 1h 07m 50s\n- openstack-ansible-deploy-infra_lxc_stepca-ubuntu-jammy https://zuul.opendev.org/t/openstack/build/a675d0f6325c4685835ac800ddeb3ee8 : FAILURE in 1h 15m 20s\n- openstack-ansible-deploy-infra_lxc_tls-ubuntu-jammy https://zuul.opendev.org/t/openstack/build/4d32bc3b7abd40e391a25f74f6a02c97 : SUCCESS in 1h 11m 57s\n- openstack-ansible-upgrade-infra_lxc-rockylinux-9 https://zuul.opendev.org/t/openstack/build/f23ccb800a344483befcc1080a391068 : SUCCESS in 1h 58m 39s\n- openstack-ansible-upgrade-infra_lxc-ubuntu-jammy https://zuul.opendev.org/t/openstack/build/38aef4b6a3114bda8de207ecc00f3aa4 : SUCCESS in 1h 34m 06s\n- openstack-ansible-deploy-hosts_lxc_zfs-ubuntu-jammy https://zuul.opendev.org/t/openstack/build/a25fad8bb7f94fbeada9b29fa1effa76 : SUCCESS in 27m 00s\n- openstack-ansible-deploy-hosts_lxc_btrfs-ubuntu-jammy https://zuul.opendev.org/t/openstack/build/1de204d1f75c4a37854a7bc439f3aad9 : SUCCESS in 31m 54s\n- openstack-ansible-deploy-hosts_lxc_overlayfs-ubuntu-jammy https://zuul.opendev.org/t/openstack/build/2ba9fce43775433fbeebdacfaf3f3e30 : SUCCESS in 26m 13s\n- openstack-ansible-deploy-hosts_lxc_btrfs-debian-bookworm https://zuul.opendev.org/t/openstack/build/b6b8e581a757452fa5c5cdd099afbe7e : SUCCESS in 26m 08s\n- openstack-ansible-deploy-hosts_lxc_overlayfs-debian-bookworm https://zuul.opendev.org/t/openstack/build/3a6f0ba456d341e0894b0800dbd2e7c0 : SUCCESS in 28m 00s\n- openstack-ansible-deploy-hosts_lxc_overlayfs-rockylinux-9 https://zuul.opendev.org/t/openstack/build/fef84d451bf244e79c148b8e8d749890 : SUCCESS in 27m 28s","accounts_in_message":[],"_revision_number":2},{"id":"c3ff87568496250dae15fb619f3e6e913e760cd7","author":{"_account_id":25023,"name":"Jonathan Rosser","email":"jonathan.rosser@rd.bbc.co.uk","username":"jrosser"},"date":"2024-07-20 07:54:09.000000000","message":"Patch Set 2:\n\n(1 comment)","accounts_in_message":[],"_revision_number":2},{"id":"d05004a05511e4efdb3e946eb87c271f148f0e7f","tag":"autogenerated:zuul:check","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2024-07-20 09:58:52.000000000","message":"Patch Set 2: Verified+1\n\nBuild succeeded (check pipeline).\nhttps://zuul.opendev.org/t/openstack/buildset/b2267ce28d4e4d2f881c2be773f6f999\n\n- openstack-ansible-linters-ubuntu-jammy https://zuul.opendev.org/t/openstack/build/822933c446f54992bc938a79ab3b92d2 : SUCCESS in 13m 16s\n- openstack-tox-docs https://zuul.opendev.org/t/openstack/build/2fd83330aa61421192cee963df172e39 : SUCCESS in 5m 58s\n- openstack-ansible-deploy-infra_lxc-centos-9-stream https://zuul.opendev.org/t/openstack/build/656d0133ef284d2682e578126da68573 : SUCCESS in 1h 13m 28s (non-voting)\n- openstack-ansible-deploy-infra_lxc-rockylinux-9 https://zuul.opendev.org/t/openstack/build/925ca771cc76448698bb73769f3ce4c2 : SUCCESS in 1h 29m 45s\n- openstack-ansible-deploy-infra_lxc_tls-rockylinux-9 https://zuul.opendev.org/t/openstack/build/8842f132955c4f4f89788e9e1fa00569 : SUCCESS in 1h 34m 29s\n- openstack-ansible-deploy-infra_lxc-debian-bookworm https://zuul.opendev.org/t/openstack/build/86de26a6a4604894bb7d8459c2caf0e6 : SUCCESS in 1h 19m 37s\n- openstack-ansible-deploy-infra_lxc-ubuntu-jammy https://zuul.opendev.org/t/openstack/build/7cbd66d13f9f4bf2b7794bb200c3df3e : SUCCESS in 1h 15m 29s\n- openstack-ansible-deploy-infra_lxc_stepca-ubuntu-jammy https://zuul.opendev.org/t/openstack/build/ae8912971a0543348be42df570c28383 : SUCCESS in 1h 07m 11s\n- openstack-ansible-deploy-infra_lxc_tls-ubuntu-jammy https://zuul.opendev.org/t/openstack/build/fbbd985743904322a18c6b468291b036 : SUCCESS in 1h 17m 33s\n- openstack-ansible-upgrade-infra_lxc-rockylinux-9 https://zuul.opendev.org/t/openstack/build/be54f0fd708146749493393a22c61832 : SUCCESS in 1h 57m 50s\n- openstack-ansible-upgrade-infra_lxc-ubuntu-jammy https://zuul.opendev.org/t/openstack/build/bb727d87dc0e4a9fa823638493b05574 : SUCCESS in 1h 32m 09s\n- openstack-ansible-deploy-hosts_lxc_zfs-ubuntu-jammy https://zuul.opendev.org/t/openstack/build/7c944c639dbd49a1bf340fc296d139f5 : SUCCESS in 28m 17s\n- openstack-ansible-deploy-hosts_lxc_btrfs-ubuntu-jammy https://zuul.opendev.org/t/openstack/build/c4db54c48b254c8190328d6a0449fc17 : SUCCESS in 33m 54s\n- openstack-ansible-deploy-hosts_lxc_overlayfs-ubuntu-jammy https://zuul.opendev.org/t/openstack/build/528e945cc2b4496098233b816046fd4d : SUCCESS in 22m 57s\n- openstack-ansible-deploy-hosts_lxc_btrfs-debian-bookworm https://zuul.opendev.org/t/openstack/build/54ae3975454d4c30a23391f88a52a418 : SUCCESS in 27m 32s\n- openstack-ansible-deploy-hosts_lxc_overlayfs-debian-bookworm https://zuul.opendev.org/t/openstack/build/fceef37c014f439799c29ce0fc4e36c6 : SUCCESS in 30m 08s\n- openstack-ansible-deploy-hosts_lxc_overlayfs-rockylinux-9 https://zuul.opendev.org/t/openstack/build/23579fe48f2345a1a4b61a2cba965c48 : SUCCESS in 27m 52s","accounts_in_message":[],"_revision_number":2},{"id":"f6a4fe75b73f306ab22a7627e564728249273435","author":{"_account_id":28619,"name":"Dmitriy Rabotyagov","email":"noonedeadpunk@gmail.com","username":"noonedeadpunk"},"date":"2024-07-22 11:08:48.000000000","message":"Patch Set 2: Code-Review+2","accounts_in_message":[],"_revision_number":2},{"id":"ab7a419b2bfa3add6fb1e61c367ac0660cadb1dd","author":{"_account_id":31542,"name":"Andrew Bonney","email":"andrew.bonney@bbc.co.uk","username":"andrewbonney"},"date":"2024-07-23 12:50:29.000000000","message":"Patch Set 2: Code-Review+2","accounts_in_message":[],"_revision_number":2},{"id":"a2c9fae3ab0fae97df1001bd180a0369334747fb","author":{"_account_id":28619,"name":"Dmitriy Rabotyagov","email":"noonedeadpunk@gmail.com","username":"noonedeadpunk"},"date":"2024-07-24 11:59:54.000000000","message":"Patch Set 2: Workflow+1","accounts_in_message":[],"_revision_number":2},{"id":"eec75506493f1cb2295602037bdc29f0541523b7","tag":"autogenerated:zuul:gate","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2024-07-24 12:01:55.000000000","message":"Patch Set 2: -Verified\n\nStarting gate jobs.","accounts_in_message":[],"_revision_number":2},{"id":"0d396cfa400fdb1aa48094f23e4a24042b77f9ca","tag":"autogenerated:zuul:gate","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2024-07-24 13:56:12.000000000","message":"Patch Set 2: Verified+2\n\nBuild succeeded (gate pipeline).\nhttps://zuul.opendev.org/t/openstack/buildset/1597f6de6f0d475a835206d3f1f0f63b\n\n- openstack-ansible-linters-ubuntu-jammy https://zuul.opendev.org/t/openstack/build/df7f3a5f1dfb4536a696d2b4e0a57b74 : SUCCESS in 9m 25s\n- openstack-tox-docs https://zuul.opendev.org/t/openstack/build/aef8b8a8f2d742e6a290115e167d468d : SUCCESS in 5m 27s\n- openstack-ansible-deploy-infra_lxc-rockylinux-9 https://zuul.opendev.org/t/openstack/build/0c8c5f8bc6664bd784b1dec36b11dd34 : SUCCESS in 1h 11m 42s\n- openstack-ansible-deploy-infra_lxc_tls-rockylinux-9 https://zuul.opendev.org/t/openstack/build/904847f7c8814fc29d7006b9560866a1 : SUCCESS in 1h 25m 21s\n- openstack-ansible-deploy-infra_lxc-debian-bookworm https://zuul.opendev.org/t/openstack/build/b1b63fcb513f458aa0c297e4c1c11953 : SUCCESS in 1h 11m 29s\n- openstack-ansible-deploy-infra_lxc-ubuntu-jammy https://zuul.opendev.org/t/openstack/build/45e6b1021eb64ad3a8491644b112b884 : SUCCESS in 1h 08m 27s\n- openstack-ansible-deploy-infra_lxc_stepca-ubuntu-jammy https://zuul.opendev.org/t/openstack/build/0bbf5fa70dff440a98434a121c87cd44 : SUCCESS in 1h 17m 23s\n- openstack-ansible-deploy-infra_lxc_tls-ubuntu-jammy https://zuul.opendev.org/t/openstack/build/142bfdac87234485a867c6a631b77cfc : SUCCESS in 1h 10m 16s\n- openstack-ansible-upgrade-infra_lxc-rockylinux-9 https://zuul.opendev.org/t/openstack/build/c539a7ad4078470da247818dd6a58184 : SUCCESS in 1h 49m 45s\n- openstack-ansible-upgrade-infra_lxc-ubuntu-jammy https://zuul.opendev.org/t/openstack/build/582dce60a20e4e9eb660d2bef2bb35e5 : SUCCESS in 1h 25m 48s\n- openstack-ansible-deploy-hosts_lxc_zfs-ubuntu-jammy https://zuul.opendev.org/t/openstack/build/531154e392b4444fbeeccc92ed15df8c : SUCCESS in 26m 15s\n- openstack-ansible-deploy-hosts_lxc_btrfs-ubuntu-jammy https://zuul.opendev.org/t/openstack/build/6b364507565445488cfa5314a5dafed3 : SUCCESS in 25m 19s\n- openstack-ansible-deploy-hosts_lxc_overlayfs-ubuntu-jammy https://zuul.opendev.org/t/openstack/build/3a90b21f280b447a93a6d452c9da4fbf : SUCCESS in 29m 54s\n- openstack-ansible-deploy-hosts_lxc_btrfs-debian-bookworm https://zuul.opendev.org/t/openstack/build/fe5304a8f77b45898f1b9b1af5217c32 : SUCCESS in 22m 49s\n- openstack-ansible-deploy-hosts_lxc_overlayfs-debian-bookworm https://zuul.opendev.org/t/openstack/build/50cd348eaf784f93b197cc879cd1c81c : SUCCESS in 25m 48s\n- openstack-ansible-deploy-hosts_lxc_overlayfs-rockylinux-9 https://zuul.opendev.org/t/openstack/build/4377ace208f34bc8b33f36a0265705a9 : SUCCESS in 26m 46s","accounts_in_message":[],"_revision_number":2},{"id":"ba1bd796303ee507f01e27b39fdf26dbc2b262f7","tag":"autogenerated:gerrit:merged","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2024-07-24 13:56:12.000000000","message":"Change has been successfully merged","accounts_in_message":[],"_revision_number":2},{"id":"480159fad1b36b7f4eccb58294985d45d465d823","tag":"autogenerated:zuul:promote","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2024-07-24 13:57:20.000000000","message":"Patch Set 2:\n\nBuild succeeded (promote pipeline).\nhttps://zuul.opendev.org/t/openstack/buildset/637f70ab5fe84d929e9bf90a12e269ec\n\n- promote-openstack-tox-docs https://zuul.opendev.org/t/openstack/build/a7ed1ff04fb942568c6645aa122a4cb5 : SUCCESS in 46s","accounts_in_message":[],"_revision_number":2}],"current_revision_number":2,"current_revision":"7b5fc5afab419afc9f17e7286375ad6b08b5d20d","revisions":{"ade4e314815f99a8bc183cb1b25ef72be21f0546":{"kind":"REWORK","_number":1,"created":"2024-07-17 15:26:05.000000000","uploader":{"_account_id":25023,"name":"Jonathan Rosser","email":"jonathan.rosser@rd.bbc.co.uk","username":"jrosser"},"ref":"refs/changes/40/924340/1","fetch":{"anonymous http":{"url":"https://review.opendev.org/openstack/openstack-ansible-lxc_hosts","ref":"refs/changes/40/924340/1","commands":{"Checkout":"git fetch https://review.opendev.org/openstack/openstack-ansible-lxc_hosts refs/changes/40/924340/1 \u0026\u0026 git checkout FETCH_HEAD","Cherry Pick":"git fetch https://review.opendev.org/openstack/openstack-ansible-lxc_hosts refs/changes/40/924340/1 \u0026\u0026 git cherry-pick FETCH_HEAD","Format Patch":"git fetch https://review.opendev.org/openstack/openstack-ansible-lxc_hosts refs/changes/40/924340/1 \u0026\u0026 git format-patch -1 --stdout FETCH_HEAD","Pull":"git pull https://review.opendev.org/openstack/openstack-ansible-lxc_hosts refs/changes/40/924340/1"}}},"commit":{"parents":[{"commit":"6984b241452007ae6c577f5759d8d342a4233ef2","subject":"Install systemd-resolved into container base image for Ubuntu Noble","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/openstack-ansible-lxc_hosts/commit/6984b241452007ae6c577f5759d8d342a4233ef2"}]}],"author":{"name":"Jonathan Rosser","email":"jonathan.rosser@rd.bbc.co.uk","date":"2024-07-17 15:25:13.000000000","tz":60},"committer":{"name":"Jonathan Rosser","email":"jonathan.rosser@rd.bbc.co.uk","date":"2024-07-17 15:25:13.000000000","tz":60},"subject":"Use generated apparmor profile by default in lxc base config","message":"Use generated apparmor profile by default in lxc base config\n\nThis is the default setting in the upstream ubuntu 24.04 lxc image.\n\nChange-Id: I7fe143f1812cd2fe3b50c9a327cec44f4523c1ab\n","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/openstack-ansible-lxc_hosts/commit/ade4e314815f99a8bc183cb1b25ef72be21f0546"}],"resolve_conflicts_web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/openstack-ansible-lxc_hosts/commit/ade4e314815f99a8bc183cb1b25ef72be21f0546"}]},"branch":"refs/heads/master"},"7b5fc5afab419afc9f17e7286375ad6b08b5d20d":{"kind":"NO_CHANGE","_number":2,"created":"2024-07-19 19:10:32.000000000","uploader":{"_account_id":25023,"name":"Jonathan Rosser","email":"jonathan.rosser@rd.bbc.co.uk","username":"jrosser"},"ref":"refs/changes/40/924340/2","fetch":{"anonymous http":{"url":"https://review.opendev.org/openstack/openstack-ansible-lxc_hosts","ref":"refs/changes/40/924340/2","commands":{"Checkout":"git fetch https://review.opendev.org/openstack/openstack-ansible-lxc_hosts refs/changes/40/924340/2 \u0026\u0026 git checkout FETCH_HEAD","Cherry Pick":"git fetch https://review.opendev.org/openstack/openstack-ansible-lxc_hosts refs/changes/40/924340/2 \u0026\u0026 git cherry-pick FETCH_HEAD","Format Patch":"git fetch https://review.opendev.org/openstack/openstack-ansible-lxc_hosts refs/changes/40/924340/2 \u0026\u0026 git format-patch -1 --stdout FETCH_HEAD","Pull":"git pull https://review.opendev.org/openstack/openstack-ansible-lxc_hosts refs/changes/40/924340/2"}}},"commit":{"parents":[{"commit":"bab5371c632749d81d7165afc830cb89fc700a34","subject":"Install systemd-resolved into container base image for Ubuntu Noble","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/openstack-ansible-lxc_hosts/commit/bab5371c632749d81d7165afc830cb89fc700a34"}]}],"author":{"name":"Jonathan Rosser","email":"jonathan.rosser@rd.bbc.co.uk","date":"2024-07-17 15:25:13.000000000","tz":60},"committer":{"name":"Jonathan Rosser","email":"jonathan.rosser@rd.bbc.co.uk","date":"2024-07-19 19:10:32.000000000","tz":0},"subject":"Use generated apparmor profile by default in lxc base config","message":"Use generated apparmor profile by default in lxc base config\n\nThis is the default setting in the upstream ubuntu 24.04 lxc image.\n\nChange-Id: I7fe143f1812cd2fe3b50c9a327cec44f4523c1ab\n","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/openstack-ansible-lxc_hosts/commit/7b5fc5afab419afc9f17e7286375ad6b08b5d20d"}],"resolve_conflicts_web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/openstack-ansible-lxc_hosts/commit/7b5fc5afab419afc9f17e7286375ad6b08b5d20d"}]},"branch":"refs/heads/master","description":"Rebase"}},"requirements":[],"submit_records":[{"rule_name":"gerrit~DefaultSubmitRule","status":"CLOSED","labels":[{"label":"Verified","status":"MAY","applied_by":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]}},{"label":"Code-Review","status":"MAY","applied_by":{"_account_id":31542,"name":"Andrew Bonney","email":"andrew.bonney@bbc.co.uk","username":"andrewbonney"}},{"label":"Workflow","status":"MAY","applied_by":{"_account_id":28619,"name":"Dmitriy Rabotyagov","email":"noonedeadpunk@gmail.com","username":"noonedeadpunk"}},{"label":"Backport-Candidate","status":"MAY"}]}],"submit_requirements":[{"name":"Verified","description":"Verified in gate by CI","status":"SATISFIED","is_legacy":false,"submittability_expression_result":{"expression":"label:Verified\u003dMAX AND -label:Verified\u003dMIN","fulfilled":true,"status":"PASS","passing_atoms":["label:Verified\u003dMAX"],"failing_atoms":["label:Verified\u003dMIN"],"atom_explanations":{}}},{"name":"Backport-Candidate","description":"Backport candidate status","status":"NOT_APPLICABLE","is_legacy":false,"applicability_expression_result":{"fulfilled":false,"status":"FAIL"},"submittability_expression_result":{"expression":"is:true","fulfilled":true,"status":"NOT_EVALUATED","passing_atoms":[],"failing_atoms":[],"atom_explanations":{}}},{"name":"Code-Review","description":"Code reviewed by core reviewer","status":"SATISFIED","is_legacy":false,"submittability_expression_result":{"expression":"label:Code-Review\u003dMAX AND -label:Code-Review\u003dMIN","fulfilled":true,"status":"PASS","passing_atoms":["label:Code-Review\u003dMAX"],"failing_atoms":["label:Code-Review\u003dMIN"],"atom_explanations":{}}},{"name":"Workflow","description":"Approved for gate by core reviewer","status":"SATISFIED","is_legacy":false,"submittability_expression_result":{"expression":"label:Workflow\u003dMAX AND -label:Workflow\u003dMIN","fulfilled":true,"status":"PASS","passing_atoms":["label:Workflow\u003dMAX"],"failing_atoms":["label:Workflow\u003dMIN"],"atom_explanations":{}}}]}
