)]}'
{"id":"openstack%2Fopenstack-ansible-lxc_hosts~939604","triplet_id":"openstack%2Fopenstack-ansible-lxc_hosts~master~I76cc0d1fe550dbb3b8c790389dd0a84ca3674729","project":"openstack/openstack-ansible-lxc_hosts","branch":"master","attention_set":{},"removed_from_attention_set":{"27316":{"account":{"_account_id":27316,"name":"Daniel Preussker","email":"openstack@devilcode.org","username":"f0o"},"last_update":"2025-02-10 11:40:41.000000000","reason":"Change was submitted"}},"hashtags":[],"change_id":"I76cc0d1fe550dbb3b8c790389dd0a84ca3674729","subject":"Make sysctl configuration path configurable Defaults to /etc/sysctl.conf to retain current behavior","status":"MERGED","created":"2025-01-19 08:14:13.000000000","updated":"2025-02-10 11:41:36.000000000","submitted":"2025-02-10 11:40:41.000000000","submitter":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"total_comment_count":1,"unresolved_comment_count":0,"has_review_started":true,"submission_id":"939604","meta_rev_id":"e97e9c2653c55a222d7c41ee3dfb73cf632965bd","_number":939604,"virtual_id_number":939604,"owner":{"_account_id":27316,"name":"Daniel Preussker","email":"openstack@devilcode.org","username":"f0o"},"actions":{},"labels":{"Verified":{"approved":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"all":[{"value":0,"_account_id":28619,"name":"Dmitriy Rabotyagov","email":"noonedeadpunk@gmail.com","username":"noonedeadpunk"},{"tag":"autogenerated:zuul:gate","value":2,"date":"2025-02-10 11:40:41.000000000","permitted_voting_range":{"min":2,"max":2},"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},{"value":0,"_account_id":25023,"name":"Jonathan Rosser","email":"jonathan.rosser@rd.bbc.co.uk","username":"jrosser"}],"values":{"-2":"Fails","-1":"Doesn\u0027t seem to work"," 0":"No score","+1":"Works for me","+2":"Verified"},"description":"","default_value":0,"optional":true},"Code-Review":{"approved":{"_account_id":28619,"name":"Dmitriy Rabotyagov","email":"noonedeadpunk@gmail.com","username":"noonedeadpunk"},"all":[{"value":2,"date":"2025-02-10 09:12:52.000000000","permitted_voting_range":{"min":2,"max":2},"_account_id":28619,"name":"Dmitriy Rabotyagov","email":"noonedeadpunk@gmail.com","username":"noonedeadpunk"},{"value":0,"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},{"value":2,"date":"2025-02-05 08:59:39.000000000","permitted_voting_range":{"min":2,"max":2},"_account_id":25023,"name":"Jonathan Rosser","email":"jonathan.rosser@rd.bbc.co.uk","username":"jrosser"}],"values":{"-2":"Do not merge","-1":"This patch needs further work before it can be merged"," 0":"No score","+1":"Looks good to me, but someone else must approve","+2":"Looks good to me (core reviewer)"},"description":"","default_value":0,"optional":true},"Workflow":{"approved":{"_account_id":28619,"name":"Dmitriy Rabotyagov","email":"noonedeadpunk@gmail.com","username":"noonedeadpunk"},"all":[{"value":1,"date":"2025-02-10 09:12:52.000000000","permitted_voting_range":{"min":1,"max":1},"_account_id":28619,"name":"Dmitriy Rabotyagov","email":"noonedeadpunk@gmail.com","username":"noonedeadpunk"},{"value":0,"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},{"value":0,"_account_id":25023,"name":"Jonathan Rosser","email":"jonathan.rosser@rd.bbc.co.uk","username":"jrosser"}],"values":{"-1":"Work in progress"," 0":"Ready for reviews","+1":"Approved"},"description":"","default_value":0,"optional":true},"Backport-Candidate":{"all":[{"value":0,"_account_id":28619,"name":"Dmitriy Rabotyagov","email":"noonedeadpunk@gmail.com","username":"noonedeadpunk"},{"value":0,"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},{"value":0,"_account_id":25023,"name":"Jonathan Rosser","email":"jonathan.rosser@rd.bbc.co.uk","username":"jrosser"}],"values":{"-1":"Not A Backport Candidate"," 0":"Backport Review Needed","+1":"Proposed Backport"},"description":"","default_value":0,"optional":true}},"removable_reviewers":[],"reviewers":{"REVIEWER":[{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},{"_account_id":25023,"name":"Jonathan Rosser","email":"jonathan.rosser@rd.bbc.co.uk","username":"jrosser"},{"_account_id":28619,"name":"Dmitriy Rabotyagov","email":"noonedeadpunk@gmail.com","username":"noonedeadpunk"}]},"pending_reviewers":{},"reviewer_updates":[{"updated":"2025-01-19 11:00:51.000000000","updated_by":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"reviewer":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"state":"REVIEWER"},{"updated":"2025-02-04 18:31:54.000000000","updated_by":{"_account_id":25023,"name":"Jonathan Rosser","email":"jonathan.rosser@rd.bbc.co.uk","username":"jrosser"},"reviewer":{"_account_id":25023,"name":"Jonathan Rosser","email":"jonathan.rosser@rd.bbc.co.uk","username":"jrosser"},"state":"CC"},{"updated":"2025-02-05 08:59:39.000000000","updated_by":{"_account_id":25023,"name":"Jonathan Rosser","email":"jonathan.rosser@rd.bbc.co.uk","username":"jrosser"},"reviewer":{"_account_id":25023,"name":"Jonathan Rosser","email":"jonathan.rosser@rd.bbc.co.uk","username":"jrosser"},"state":"REVIEWER"},{"updated":"2025-02-10 09:12:52.000000000","updated_by":{"_account_id":28619,"name":"Dmitriy Rabotyagov","email":"noonedeadpunk@gmail.com","username":"noonedeadpunk"},"reviewer":{"_account_id":28619,"name":"Dmitriy Rabotyagov","email":"noonedeadpunk@gmail.com","username":"noonedeadpunk"},"state":"REVIEWER"}],"messages":[{"id":"857fb6f7d503399ff911ac667068b94ab596db00","tag":"autogenerated:gerrit:newPatchSet","author":{"_account_id":27316,"name":"Daniel Preussker","email":"openstack@devilcode.org","username":"f0o"},"date":"2025-01-19 08:14:13.000000000","message":"Uploaded patch set 1.","accounts_in_message":[],"_revision_number":1},{"id":"1b3e451045ce4781ab4c4bc27025e9193003c765","tag":"autogenerated:gerrit:newPatchSet","author":{"_account_id":27316,"name":"Daniel Preussker","email":"openstack@devilcode.org","username":"f0o"},"date":"2025-01-19 09:01:03.000000000","message":"Uploaded patch set 2.","accounts_in_message":[],"_revision_number":2},{"id":"53c68880b71a681fdddab54ad1104ff94d6ba526","tag":"autogenerated:zuul:check","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2025-01-19 11:00:51.000000000","message":"Patch Set 2: Verified-1\n\nBuild failed (check pipeline).  For information on how to proceed, see\nhttps://docs.opendev.org/opendev/infra-manual/latest/developers.html#automated-testing\nand https://docs.openstack.org/project-team-guide/testing.html#how-to-handle-test-failures\n\nhttps://zuul.opendev.org/t/openstack/buildset/50b4c195611b4f679946d4425081d81a\n\n- openstack-ansible-linters-ubuntu-noble https://zuul.opendev.org/t/openstack/build/ce33ac1e7c14467fb4cebe42408a0dac : SUCCESS in 5m 40s\n- openstack-tox-docs https://zuul.opendev.org/t/openstack/build/e848de4a908944b4b999e16436c87e25 : SUCCESS in 3m 04s\n- openstack-ansible-deploy-infra_lxc-centos-9-stream https://zuul.opendev.org/t/openstack/build/67cb927869374821b8f0db3afbba5a12 : SUCCESS in 57m 21s (non-voting)\n- openstack-ansible-deploy-infra_lxc-rockylinux-9 https://zuul.opendev.org/t/openstack/build/72428850bd1644bc927c1e764627fa79 : SUCCESS in 1h 24m 29s\n- openstack-ansible-deploy-infra_lxc_tls-rockylinux-9 https://zuul.opendev.org/t/openstack/build/6ae41827c4aa40a2afed54820e3eee25 : SUCCESS in 1h 02m 59s\n- openstack-ansible-deploy-infra_lxc-debian-bookworm https://zuul.opendev.org/t/openstack/build/ab6681f93fb2400ebca50099e9a40b39 : SUCCESS in 1h 17m 37s\n- openstack-ansible-deploy-infra_lxc-ubuntu-jammy https://zuul.opendev.org/t/openstack/build/502cd25d4e604ac7977964b0fa7640f1 : SUCCESS in 1h 10m 44s\n- openstack-ansible-deploy-infra_lxc_stepca-ubuntu-jammy https://zuul.opendev.org/t/openstack/build/d2d6ea2d27484ad89399d36ca34fbc12 : SUCCESS in 1h 06m 21s\n- openstack-ansible-deploy-infra_lxc_tls-ubuntu-jammy https://zuul.opendev.org/t/openstack/build/89e8a43742b74df6877f96932aa98197 : SUCCESS in 1h 18m 02s\n- openstack-ansible-deploy-infra_lxc-ubuntu-noble https://zuul.opendev.org/t/openstack/build/c1eb89e8911844d0b8902bf144960744 : SUCCESS in 1h 02m 53s\n- openstack-ansible-deploy-infra_lxc_stepca-ubuntu-noble https://zuul.opendev.org/t/openstack/build/baf79e03bb1e47808523059cfaab5278 : SUCCESS in 1h 14m 03s\n- openstack-ansible-deploy-infra_lxc_tls-ubuntu-noble https://zuul.opendev.org/t/openstack/build/b3e853b0f93346d3ae2a55394c29e5dc : SUCCESS in 1h 19m 05s\n- openstack-ansible-upgrade-infra_lxc-rockylinux-9 https://zuul.opendev.org/t/openstack/build/a004b0503016461dad121b381a4c7012 : SUCCESS in 1h 55m 13s\n- openstack-ansible-upgrade_2024.1-infra_lxc-rockylinux-9 https://zuul.opendev.org/t/openstack/build/6c9dac20270d46e8bc2ffcf39cc15d41 : SUCCESS in 1h 27m 27s\n- openstack-ansible-upgrade-infra_lxc-ubuntu-jammy https://zuul.opendev.org/t/openstack/build/8fdc700f444b4cc09e6b47d433522047 : SUCCESS in 1h 28m 27s\n- openstack-ansible-upgrade_2024.1-infra_lxc-ubuntu-jammy https://zuul.opendev.org/t/openstack/build/0803b0af4a5f45aa96e28868ec056c4e : FAILURE in 42m 09s\n- openstack-ansible-deploy-hosts_lxc_zfs-ubuntu-jammy https://zuul.opendev.org/t/openstack/build/d2d97524a4804df4a8fd899dde534523 : SUCCESS in 20m 27s\n- openstack-ansible-deploy-hosts_lxc_btrfs-ubuntu-jammy https://zuul.opendev.org/t/openstack/build/95149b43bcf04222969d6d294026b7be : SUCCESS in 23m 23s\n- openstack-ansible-deploy-hosts_lxc_overlayfs-ubuntu-jammy https://zuul.opendev.org/t/openstack/build/608dcd6953b544c9b3bf43a23adfac65 : SUCCESS in 19m 24s\n- openstack-ansible-deploy-hosts_lxc_zfs-ubuntu-noble https://zuul.opendev.org/t/openstack/build/27b5c593fd754af495f19b819a9f4e67 : SUCCESS in 22m 53s\n- openstack-ansible-deploy-hosts_lxc_btrfs-ubuntu-noble https://zuul.opendev.org/t/openstack/build/3cf3a4a260e941fa8182e3ad0f73a9b0 : SUCCESS in 24m 17s\n- openstack-ansible-deploy-hosts_lxc_overlayfs-ubuntu-noble https://zuul.opendev.org/t/openstack/build/9fe07c0acc4441c88fc4822abc5930ca : SUCCESS in 18m 59s\n- openstack-ansible-deploy-hosts_lxc_btrfs-debian-bookworm https://zuul.opendev.org/t/openstack/build/40f7c96ef7c944c28b6da94d0d2d235e : SUCCESS in 20m 31s\n- openstack-ansible-deploy-hosts_lxc_overlayfs-debian-bookworm https://zuul.opendev.org/t/openstack/build/be0a625af74947b6bfc7f32cae4f7d10 : SUCCESS in 23m 56s\n- openstack-ansible-deploy-hosts_lxc_overlayfs-rockylinux-9 https://zuul.opendev.org/t/openstack/build/ad540d1520e34c8c985dc64e07993e29 : SUCCESS in 20m 12s","accounts_in_message":[],"_revision_number":2},{"id":"0f8d15305682533664b42446785c4fb177752295","author":{"_account_id":25023,"name":"Jonathan Rosser","email":"jonathan.rosser@rd.bbc.co.uk","username":"jrosser"},"date":"2025-02-04 18:31:54.000000000","message":"Patch Set 2:\n\n(1 comment)","accounts_in_message":[],"_revision_number":2},{"id":"516a22876c06fd389a8d5ff193e61a1017045e6c","tag":"autogenerated:zuul:check","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2025-02-04 21:10:20.000000000","message":"Patch Set 2: Verified+1\n\nBuild succeeded (check pipeline).\nhttps://zuul.opendev.org/t/openstack/buildset/fe9b6b860dda4dcab0859453b0836eb8\n\n- openstack-ansible-linters-ubuntu-noble https://zuul.opendev.org/t/openstack/build/ebc7ba32db4c47a486706a0789cfc8bf : SUCCESS in 4m 59s\n- openstack-tox-docs https://zuul.opendev.org/t/openstack/build/0d3fab0880614bfca4e996e37b38b695 : SUCCESS in 5m 00s\n- openstack-ansible-deploy-infra_lxc-centos-9-stream https://zuul.opendev.org/t/openstack/build/595fc69dc9364650bcaaa0bbcb957f5b : SUCCESS in 1h 17m 17s (non-voting)\n- openstack-ansible-deploy-infra_lxc-rockylinux-9 https://zuul.opendev.org/t/openstack/build/aeed2c208ebc40c1b7082910bb3dec1d : SUCCESS in 1h 31m 13s\n- openstack-ansible-deploy-infra_lxc_tls-rockylinux-9 https://zuul.opendev.org/t/openstack/build/627c66d897ff423fa93eeea36571c97d : SUCCESS in 1h 15m 07s\n- openstack-ansible-deploy-infra_lxc-debian-bookworm https://zuul.opendev.org/t/openstack/build/a320115efd5b4861a2c320161ff843fb : SUCCESS in 1h 07m 48s\n- openstack-ansible-deploy-infra_lxc-ubuntu-jammy https://zuul.opendev.org/t/openstack/build/5042060567a24bfc9f6588606b87aadc : SUCCESS in 43m 03s\n- openstack-ansible-deploy-infra_lxc_stepca-ubuntu-jammy https://zuul.opendev.org/t/openstack/build/96ac294c27fe47669f7f0291fcc9f59d : SUCCESS in 1h 10m 26s\n- openstack-ansible-deploy-infra_lxc_tls-ubuntu-jammy https://zuul.opendev.org/t/openstack/build/8a9f7f8b7bcc4bc296f880ebb2bba7f1 : SUCCESS in 1h 16m 22s\n- openstack-ansible-deploy-infra_lxc-ubuntu-noble https://zuul.opendev.org/t/openstack/build/f27f6556c2e34283b4c7044d1b10b0c5 : SUCCESS in 1h 32m 44s\n- openstack-ansible-deploy-infra_lxc_stepca-ubuntu-noble https://zuul.opendev.org/t/openstack/build/e11726d653ec4d9ebdece1bf1717783b : SUCCESS in 1h 18m 09s\n- openstack-ansible-deploy-infra_lxc_tls-ubuntu-noble https://zuul.opendev.org/t/openstack/build/fe859e95a240457d8b9d56c40278ae59 : SUCCESS in 1h 15m 02s\n- openstack-ansible-upgrade-infra_lxc-rockylinux-9 https://zuul.opendev.org/t/openstack/build/1dc3b5c9a7364bada5e76344d985097b : SUCCESS in 1h 59m 41s\n- openstack-ansible-upgrade_2024.1-infra_lxc-rockylinux-9 https://zuul.opendev.org/t/openstack/build/bd09438229064c3d86e7f4a63325f04c : SUCCESS in 1h 56m 14s\n- openstack-ansible-upgrade-infra_lxc-ubuntu-jammy https://zuul.opendev.org/t/openstack/build/456ad7a7a12c491da7987709191dad84 : SUCCESS in 1h 38m 36s\n- openstack-ansible-upgrade_2024.1-infra_lxc-ubuntu-jammy https://zuul.opendev.org/t/openstack/build/cae6899cd36a426a87ff3b48ad51ac70 : SUCCESS in 1h 37m 35s\n- openstack-ansible-deploy-hosts_lxc_zfs-ubuntu-jammy https://zuul.opendev.org/t/openstack/build/44bd8dd32ed848838a12515cb41eaaf1 : SUCCESS in 19m 37s\n- openstack-ansible-deploy-hosts_lxc_btrfs-ubuntu-jammy https://zuul.opendev.org/t/openstack/build/c11244bcc9a34ae6aea9bab30556af8b : SUCCESS in 20m 36s\n- openstack-ansible-deploy-hosts_lxc_overlayfs-ubuntu-jammy https://zuul.opendev.org/t/openstack/build/62b5de4376fd4ed0be7eb7adcfcd8b1b : SUCCESS in 22m 23s\n- openstack-ansible-deploy-hosts_lxc_zfs-ubuntu-noble https://zuul.opendev.org/t/openstack/build/1945aa8399fc4228b4fa4b7772556086 : SUCCESS in 25m 25s\n- openstack-ansible-deploy-hosts_lxc_btrfs-ubuntu-noble https://zuul.opendev.org/t/openstack/build/89739fa983694077acca741c5795c770 : SUCCESS in 24m 20s\n- openstack-ansible-deploy-hosts_lxc_overlayfs-ubuntu-noble https://zuul.opendev.org/t/openstack/build/acdffca6ef4844728924b6d31e0a3cdf : SUCCESS in 15m 16s\n- openstack-ansible-deploy-hosts_lxc_btrfs-debian-bookworm https://zuul.opendev.org/t/openstack/build/403eb9eae57e46cdb64133b970469c32 : SUCCESS in 21m 08s\n- openstack-ansible-deploy-hosts_lxc_overlayfs-debian-bookworm https://zuul.opendev.org/t/openstack/build/c743aae249dc4050b44e834960719c68 : SUCCESS in 24m 06s\n- openstack-ansible-deploy-hosts_lxc_overlayfs-rockylinux-9 https://zuul.opendev.org/t/openstack/build/41e0cf4a990e4095984a0f578c66bbea : SUCCESS in 20m 53s","accounts_in_message":[],"_revision_number":2},{"id":"b7708b6d5d0b0fb180a9b6e8784d89126c35e136","author":{"_account_id":25023,"name":"Jonathan Rosser","email":"jonathan.rosser@rd.bbc.co.uk","username":"jrosser"},"date":"2025-02-05 08:59:39.000000000","message":"Patch Set 2: Code-Review+2","accounts_in_message":[],"_revision_number":2},{"id":"ce7130ae6c32db627ac4d3539fd162b3e942675c","author":{"_account_id":28619,"name":"Dmitriy Rabotyagov","email":"noonedeadpunk@gmail.com","username":"noonedeadpunk"},"date":"2025-02-10 09:12:52.000000000","message":"Patch Set 2: Code-Review+2 Workflow+1","accounts_in_message":[],"_revision_number":2},{"id":"885b58aa32baf1e9e2a7a35d1408c2ac5913d859","tag":"autogenerated:zuul:gate","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2025-02-10 09:15:02.000000000","message":"Patch Set 2: -Verified\n\nStarting gate jobs.","accounts_in_message":[],"_revision_number":2},{"id":"2cba2781deb2b9127b2a3c58e16d6281ec0d1398","tag":"autogenerated:zuul:gate","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2025-02-10 11:40:41.000000000","message":"Patch Set 2: Verified+2\n\nBuild succeeded (gate pipeline).\nhttps://zuul.opendev.org/t/openstack/buildset/489a3e3996cb4663af0697ade9a079ee\n\n- openstack-ansible-linters-ubuntu-noble https://zuul.opendev.org/t/openstack/build/01484eff092b483094bf19f463b803f2 : SUCCESS in 6m 44s\n- openstack-tox-docs https://zuul.opendev.org/t/openstack/build/a0d217e7d59e480688b516b2fc49b01b : SUCCESS in 4m 13s\n- openstack-ansible-deploy-infra_lxc-rockylinux-9 https://zuul.opendev.org/t/openstack/build/6be795c21b7b4c3d8f0825b475c3b4a9 : SUCCESS in 1h 15m 50s\n- openstack-ansible-deploy-infra_lxc_tls-rockylinux-9 https://zuul.opendev.org/t/openstack/build/dec1fd6287724a25a7dda73ebc5d6fc1 : SUCCESS in 1h 29m 56s\n- openstack-ansible-deploy-infra_lxc-debian-bookworm https://zuul.opendev.org/t/openstack/build/bb29cc51035541d5affabfd3bc3b034d : SUCCESS in 1h 12m 11s\n- openstack-ansible-deploy-infra_lxc-ubuntu-jammy https://zuul.opendev.org/t/openstack/build/1a86dff4878c402eb268663e4f76558f : SUCCESS in 1h 09m 13s\n- openstack-ansible-deploy-infra_lxc_stepca-ubuntu-jammy https://zuul.opendev.org/t/openstack/build/5504d6b809d743579767faae80882573 : SUCCESS in 1h 09m 34s\n- openstack-ansible-deploy-infra_lxc_tls-ubuntu-jammy https://zuul.opendev.org/t/openstack/build/62b7eb4f962f47d89783170c9cf6aea2 : SUCCESS in 1h 14m 34s\n- openstack-ansible-deploy-infra_lxc-ubuntu-noble https://zuul.opendev.org/t/openstack/build/b08eb3ded17e4eae922004d280a7f2cf : SUCCESS in 1h 23m 26s\n- openstack-ansible-deploy-infra_lxc_stepca-ubuntu-noble https://zuul.opendev.org/t/openstack/build/f378f88dc7d64f929140d6342402c883 : SUCCESS in 1h 12m 39s\n- openstack-ansible-deploy-infra_lxc_tls-ubuntu-noble https://zuul.opendev.org/t/openstack/build/f5d225cb7eff4ae3bb66195e8cad6ef4 : SUCCESS in 1h 20m 57s\n- openstack-ansible-upgrade-infra_lxc-rockylinux-9 https://zuul.opendev.org/t/openstack/build/0d180606abee4fe3bac8ac37d6b8364c : SUCCESS in 1h 44m 08s\n- openstack-ansible-upgrade_2024.1-infra_lxc-rockylinux-9 https://zuul.opendev.org/t/openstack/build/fe14956cd19445c182911f9dc90d9f0c : SUCCESS in 1h 55m 34s\n- openstack-ansible-upgrade-infra_lxc-ubuntu-jammy https://zuul.opendev.org/t/openstack/build/eeb87007f45e4170bc4143081fb6d66e : SUCCESS in 1h 38m 47s\n- openstack-ansible-upgrade_2024.1-infra_lxc-ubuntu-jammy https://zuul.opendev.org/t/openstack/build/5a22ad9a62454172a879d52b598b5d53 : SUCCESS in 1h 27m 15s\n- openstack-ansible-deploy-hosts_lxc_zfs-ubuntu-jammy https://zuul.opendev.org/t/openstack/build/c719bcf68bd542a998474d391e82dad9 : SUCCESS in 20m 03s\n- openstack-ansible-deploy-hosts_lxc_btrfs-ubuntu-jammy https://zuul.opendev.org/t/openstack/build/8726d4fa482c44519e4205485beeb86e : SUCCESS in 20m 29s\n- openstack-ansible-deploy-hosts_lxc_overlayfs-ubuntu-jammy https://zuul.opendev.org/t/openstack/build/6a1c260126314f6e91a449947167db50 : SUCCESS in 21m 10s\n- openstack-ansible-deploy-hosts_lxc_zfs-ubuntu-noble https://zuul.opendev.org/t/openstack/build/29d7979316b649a083e1aa61ef742097 : SUCCESS in 19m 23s\n- openstack-ansible-deploy-hosts_lxc_btrfs-ubuntu-noble https://zuul.opendev.org/t/openstack/build/ac86a2c78a954ab0af1ece058fc98d55 : SUCCESS in 27m 08s\n- openstack-ansible-deploy-hosts_lxc_overlayfs-ubuntu-noble https://zuul.opendev.org/t/openstack/build/bf86ab4c259c478da2c572b9186bed37 : SUCCESS in 24m 36s\n- openstack-ansible-deploy-hosts_lxc_btrfs-debian-bookworm https://zuul.opendev.org/t/openstack/build/c63faea71c8b43469b6ec2005db36107 : SUCCESS in 21m 06s\n- openstack-ansible-deploy-hosts_lxc_overlayfs-debian-bookworm https://zuul.opendev.org/t/openstack/build/17934d0585204689bb3107f35b4ae735 : SUCCESS in 24m 15s\n- openstack-ansible-deploy-hosts_lxc_overlayfs-rockylinux-9 https://zuul.opendev.org/t/openstack/build/4cca77fb664144d7816629ea92cfa6ec : SUCCESS in 24m 07s","accounts_in_message":[],"_revision_number":2},{"id":"7bd45d2470b70967e24eafb10fd8e3ac70ac86b4","tag":"autogenerated:gerrit:merged","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2025-02-10 11:40:41.000000000","message":"Change has been successfully merged","accounts_in_message":[],"_revision_number":2},{"id":"e97e9c2653c55a222d7c41ee3dfb73cf632965bd","tag":"autogenerated:zuul:promote","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2025-02-10 11:41:36.000000000","message":"Patch Set 2:\n\nBuild succeeded (promote pipeline).\nhttps://zuul.opendev.org/t/openstack/buildset/9b8a1cd9994646038a84ece5700c384f\n\n- promote-openstack-tox-docs https://zuul.opendev.org/t/openstack/build/e6fb8fe137ca4c7f90183a3dac0f9fe0 : SUCCESS in 37s","accounts_in_message":[],"_revision_number":2}],"current_revision_number":2,"current_revision":"22b605abe401e18dfd7990d6fab10fb7b98ed7fb","revisions":{"9b53db8de8aafa690cf9278b65a2fc6779e95075":{"kind":"REWORK","_number":1,"created":"2025-01-19 08:14:13.000000000","uploader":{"_account_id":27316,"name":"Daniel Preussker","email":"openstack@devilcode.org","username":"f0o"},"ref":"refs/changes/04/939604/1","fetch":{"anonymous http":{"url":"https://review.opendev.org/openstack/openstack-ansible-lxc_hosts","ref":"refs/changes/04/939604/1","commands":{"Checkout":"git fetch https://review.opendev.org/openstack/openstack-ansible-lxc_hosts refs/changes/04/939604/1 \u0026\u0026 git checkout FETCH_HEAD","Cherry Pick":"git fetch https://review.opendev.org/openstack/openstack-ansible-lxc_hosts refs/changes/04/939604/1 \u0026\u0026 git cherry-pick FETCH_HEAD","Format Patch":"git fetch https://review.opendev.org/openstack/openstack-ansible-lxc_hosts refs/changes/04/939604/1 \u0026\u0026 git format-patch -1 --stdout FETCH_HEAD","Pull":"git pull https://review.opendev.org/openstack/openstack-ansible-lxc_hosts refs/changes/04/939604/1"}}},"commit":{"parents":[{"commit":"b1235fbfc6a7959cd698f445a2027c75a063a1dc","subject":"Merge \"Delete centos-private-epel release note\"","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/openstack-ansible-lxc_hosts/commit/b1235fbfc6a7959cd698f445a2027c75a063a1dc"}]}],"author":{"name":"Daniel \u0027f0o\u0027 Preussker","email":"f0o@devilcode.org","date":"2025-01-19 08:14:06.000000000","tz":0},"committer":{"name":"Daniel \u0027f0o\u0027 Preussker","email":"f0o@devilcode.org","date":"2025-01-19 08:14:06.000000000","tz":0},"subject":"Make sysctl configuration path configurable Defaults to /etc/sysctl.conf to retain current behavior","message":"Make sysctl configuration path configurable\nDefaults to /etc/sysctl.conf to retain current behavior\n\nChange-Id: I76cc0d1fe550dbb3b8c790389dd0a84ca3674729\n","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/openstack-ansible-lxc_hosts/commit/9b53db8de8aafa690cf9278b65a2fc6779e95075"}],"resolve_conflicts_web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/openstack-ansible-lxc_hosts/commit/9b53db8de8aafa690cf9278b65a2fc6779e95075"}]},"branch":"refs/heads/master"},"22b605abe401e18dfd7990d6fab10fb7b98ed7fb":{"kind":"REWORK","_number":2,"created":"2025-01-19 09:01:03.000000000","uploader":{"_account_id":27316,"name":"Daniel Preussker","email":"openstack@devilcode.org","username":"f0o"},"ref":"refs/changes/04/939604/2","fetch":{"anonymous http":{"url":"https://review.opendev.org/openstack/openstack-ansible-lxc_hosts","ref":"refs/changes/04/939604/2","commands":{"Checkout":"git fetch https://review.opendev.org/openstack/openstack-ansible-lxc_hosts refs/changes/04/939604/2 \u0026\u0026 git checkout FETCH_HEAD","Cherry Pick":"git fetch https://review.opendev.org/openstack/openstack-ansible-lxc_hosts refs/changes/04/939604/2 \u0026\u0026 git cherry-pick FETCH_HEAD","Format Patch":"git fetch https://review.opendev.org/openstack/openstack-ansible-lxc_hosts refs/changes/04/939604/2 \u0026\u0026 git format-patch -1 --stdout FETCH_HEAD","Pull":"git pull https://review.opendev.org/openstack/openstack-ansible-lxc_hosts refs/changes/04/939604/2"}}},"commit":{"parents":[{"commit":"b1235fbfc6a7959cd698f445a2027c75a063a1dc","subject":"Merge \"Delete centos-private-epel release note\"","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/openstack-ansible-lxc_hosts/commit/b1235fbfc6a7959cd698f445a2027c75a063a1dc"}]}],"author":{"name":"Daniel \u0027f0o\u0027 Preussker","email":"f0o@devilcode.org","date":"2025-01-19 08:14:06.000000000","tz":0},"committer":{"name":"Daniel \u0027f0o\u0027 Preussker","email":"f0o@devilcode.org","date":"2025-01-19 09:00:57.000000000","tz":0},"subject":"Make sysctl configuration path configurable Defaults to /etc/sysctl.conf to retain current behavior","message":"Make sysctl configuration path configurable\nDefaults to /etc/sysctl.conf to retain current behavior\n\nChange-Id: I76cc0d1fe550dbb3b8c790389dd0a84ca3674729\n","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/openstack-ansible-lxc_hosts/commit/22b605abe401e18dfd7990d6fab10fb7b98ed7fb"}],"resolve_conflicts_web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/openstack-ansible-lxc_hosts/commit/22b605abe401e18dfd7990d6fab10fb7b98ed7fb"}]},"branch":"refs/heads/master"}},"requirements":[],"submit_records":[{"rule_name":"gerrit~DefaultSubmitRule","status":"CLOSED","labels":[{"label":"Verified","status":"MAY","applied_by":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]}},{"label":"Code-Review","status":"MAY","applied_by":{"_account_id":28619,"name":"Dmitriy Rabotyagov","email":"noonedeadpunk@gmail.com","username":"noonedeadpunk"}},{"label":"Workflow","status":"MAY","applied_by":{"_account_id":28619,"name":"Dmitriy Rabotyagov","email":"noonedeadpunk@gmail.com","username":"noonedeadpunk"}},{"label":"Backport-Candidate","status":"MAY"}]}],"submit_requirements":[{"name":"Verified","description":"Verified in gate by CI","status":"SATISFIED","is_legacy":false,"submittability_expression_result":{"expression":"label:Verified\u003dMAX AND -label:Verified\u003dMIN","fulfilled":true,"status":"PASS","passing_atoms":["label:Verified\u003dMAX"],"failing_atoms":["label:Verified\u003dMIN"],"atom_explanations":{}}},{"name":"Backport-Candidate","description":"Backport candidate status","status":"NOT_APPLICABLE","is_legacy":false,"applicability_expression_result":{"fulfilled":false,"status":"FAIL"},"submittability_expression_result":{"expression":"is:true","fulfilled":true,"status":"NOT_EVALUATED","passing_atoms":[],"failing_atoms":[],"atom_explanations":{}}},{"name":"Code-Review","description":"Code reviewed by core reviewer","status":"SATISFIED","is_legacy":false,"submittability_expression_result":{"expression":"label:Code-Review\u003dMAX AND -label:Code-Review\u003dMIN","fulfilled":true,"status":"PASS","passing_atoms":["label:Code-Review\u003dMAX"],"failing_atoms":["label:Code-Review\u003dMIN"],"atom_explanations":{}}},{"name":"Workflow","description":"Approved for gate by core reviewer","status":"SATISFIED","is_legacy":false,"submittability_expression_result":{"expression":"label:Workflow\u003dMAX AND -label:Workflow\u003dMIN","fulfilled":true,"status":"PASS","passing_atoms":["label:Workflow\u003dMAX"],"failing_atoms":["label:Workflow\u003dMIN"],"atom_explanations":{}}}]}
