)]}'
{"id":"openstack%2Fopenstack-ansible-lxc_hosts~941528","triplet_id":"openstack%2Fopenstack-ansible-lxc_hosts~master~I84f2efb55945e0379f44fca8ddb7b92241c1d9ef","project":"openstack/openstack-ansible-lxc_hosts","branch":"master","topic":"osa/lint_autofix","attention_set":{},"removed_from_attention_set":{"28619":{"account":{"_account_id":28619,"name":"Dmitriy Rabotyagov","email":"noonedeadpunk@gmail.com","username":"noonedeadpunk"},"last_update":"2025-02-20 13:50:17.000000000","reason":"\u003cGERRIT_ACCOUNT_28619\u003e replied on the change","reason_account":{"_account_id":28619,"name":"Dmitriy Rabotyagov","email":"noonedeadpunk@gmail.com","username":"noonedeadpunk"}}},"hashtags":[],"change_id":"I84f2efb55945e0379f44fca8ddb7b92241c1d9ef","subject":"Auto-fix yaml rules","status":"MERGED","created":"2025-02-13 08:36:25.000000000","updated":"2025-02-20 18:15:06.000000000","submitted":"2025-02-20 18:14:04.000000000","submitter":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"total_comment_count":8,"unresolved_comment_count":0,"has_review_started":true,"submission_id":"941528-osa/lint_autofix","meta_rev_id":"bea23d6c765a91442506936966b24d2315944cf3","_number":941528,"virtual_id_number":941528,"owner":{"_account_id":28619,"name":"Dmitriy Rabotyagov","email":"noonedeadpunk@gmail.com","username":"noonedeadpunk"},"actions":{},"labels":{"Verified":{"approved":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"all":[{"value":0,"_account_id":31542,"name":"Andrew Bonney","email":"andrew.bonney@bbc.co.uk","username":"andrewbonney"},{"tag":"autogenerated:zuul:gate","value":2,"date":"2025-02-20 18:14:04.000000000","permitted_voting_range":{"min":2,"max":2},"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},{"value":0,"_account_id":25023,"name":"Jonathan Rosser","email":"jonathan.rosser@rd.bbc.co.uk","username":"jrosser"}],"values":{"-2":"Fails","-1":"Doesn\u0027t seem to work"," 0":"No score","+1":"Works for me","+2":"Verified"},"description":"","default_value":0,"optional":true},"Code-Review":{"approved":{"_account_id":31542,"name":"Andrew Bonney","email":"andrew.bonney@bbc.co.uk","username":"andrewbonney"},"all":[{"value":2,"date":"2025-02-19 09:07:51.000000000","permitted_voting_range":{"min":2,"max":2},"_account_id":31542,"name":"Andrew Bonney","email":"andrew.bonney@bbc.co.uk","username":"andrewbonney"},{"value":0,"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},{"value":2,"date":"2025-02-19 17:09:53.000000000","permitted_voting_range":{"min":2,"max":2},"_account_id":25023,"name":"Jonathan Rosser","email":"jonathan.rosser@rd.bbc.co.uk","username":"jrosser"}],"values":{"-2":"Do not merge","-1":"This patch needs further work before it can be merged"," 0":"No score","+1":"Looks good to me, but someone else must approve","+2":"Looks good to me (core reviewer)"},"description":"","default_value":0,"optional":true},"Workflow":{"approved":{"_account_id":25023,"name":"Jonathan Rosser","email":"jonathan.rosser@rd.bbc.co.uk","username":"jrosser"},"all":[{"value":0,"_account_id":31542,"name":"Andrew Bonney","email":"andrew.bonney@bbc.co.uk","username":"andrewbonney"},{"value":0,"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},{"value":1,"date":"2025-02-19 17:09:53.000000000","permitted_voting_range":{"min":1,"max":1},"_account_id":25023,"name":"Jonathan Rosser","email":"jonathan.rosser@rd.bbc.co.uk","username":"jrosser"}],"values":{"-1":"Work in progress"," 0":"Ready for reviews","+1":"Approved"},"description":"","default_value":0,"optional":true},"Backport-Candidate":{"all":[{"value":0,"_account_id":31542,"name":"Andrew Bonney","email":"andrew.bonney@bbc.co.uk","username":"andrewbonney"},{"value":0,"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},{"value":0,"_account_id":25023,"name":"Jonathan Rosser","email":"jonathan.rosser@rd.bbc.co.uk","username":"jrosser"}],"values":{"-1":"Not A Backport Candidate"," 0":"Backport Review Needed","+1":"Proposed Backport"},"description":"","default_value":0,"optional":true}},"removable_reviewers":[],"reviewers":{"REVIEWER":[{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},{"_account_id":25023,"name":"Jonathan Rosser","email":"jonathan.rosser@rd.bbc.co.uk","username":"jrosser"},{"_account_id":31542,"name":"Andrew Bonney","email":"andrew.bonney@bbc.co.uk","username":"andrewbonney"}]},"pending_reviewers":{},"reviewer_updates":[{"updated":"2025-02-13 10:30:17.000000000","updated_by":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"reviewer":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"state":"REVIEWER"},{"updated":"2025-02-19 08:14:25.000000000","updated_by":{"_account_id":31542,"name":"Andrew Bonney","email":"andrew.bonney@bbc.co.uk","username":"andrewbonney"},"reviewer":{"_account_id":31542,"name":"Andrew Bonney","email":"andrew.bonney@bbc.co.uk","username":"andrewbonney"},"state":"REVIEWER"},{"updated":"2025-02-19 17:09:53.000000000","updated_by":{"_account_id":25023,"name":"Jonathan Rosser","email":"jonathan.rosser@rd.bbc.co.uk","username":"jrosser"},"reviewer":{"_account_id":25023,"name":"Jonathan Rosser","email":"jonathan.rosser@rd.bbc.co.uk","username":"jrosser"},"state":"REVIEWER"}],"messages":[{"id":"0287d43ff0703da84a0a4d2ae362b7d1c30d1145","tag":"autogenerated:gerrit:newPatchSet","author":{"_account_id":28619,"name":"Dmitriy Rabotyagov","email":"noonedeadpunk@gmail.com","username":"noonedeadpunk"},"date":"2025-02-13 08:36:25.000000000","message":"Uploaded patch set 1.","accounts_in_message":[],"_revision_number":1},{"id":"95f8056b0eed901a7baf54e76b8a83dc26958660","tag":"autogenerated:zuul:check","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2025-02-13 10:30:17.000000000","message":"Patch Set 1: Verified-1\n\nBuild failed (check pipeline).  For information on how to proceed, see\nhttps://docs.opendev.org/opendev/infra-manual/latest/developers.html#automated-testing\nand https://docs.openstack.org/project-team-guide/testing.html#how-to-handle-test-failures\n\nhttps://zuul.opendev.org/t/openstack/buildset/14e7f11da9184539a59902ef0bc8d3de\n\n- openstack-ansible-linters-ubuntu-noble https://zuul.opendev.org/t/openstack/build/80a28c4981044ddb877462478f1c0539 : SUCCESS in 6m 56s\n- openstack-tox-docs https://zuul.opendev.org/t/openstack/build/5d12c7596911440c962e1321160b4927 : SUCCESS in 4m 47s\n- openstack-ansible-deploy-infra_lxc-centos-9-stream https://zuul.opendev.org/t/openstack/build/dd27f2a2125f4355bc653b91dc9c4315 : SUCCESS in 1h 22m 52s (non-voting)\n- openstack-ansible-deploy-infra_lxc-rockylinux-9 https://zuul.opendev.org/t/openstack/build/14a50a34d432441190feab71ede4b4d8 : SUCCESS in 1h 17m 03s\n- openstack-ansible-deploy-infra_lxc_tls-rockylinux-9 https://zuul.opendev.org/t/openstack/build/ddc80ccd6d6a407dbe33cce05e56c3ca : SUCCESS in 1h 26m 36s\n- openstack-ansible-deploy-infra_lxc-debian-bookworm https://zuul.opendev.org/t/openstack/build/00e7954fd8014715af6be3c6a2454a21 : SUCCESS in 1h 18m 04s\n- openstack-ansible-deploy-infra_lxc-ubuntu-jammy https://zuul.opendev.org/t/openstack/build/411394239b004dcab3269fe068a8851d : FAILURE in 1h 08m 03s\n- openstack-ansible-deploy-infra_lxc_stepca-ubuntu-jammy https://zuul.opendev.org/t/openstack/build/55b855a46f5d4e5d83199592b42aebf9 : SUCCESS in 1h 19m 20s\n- openstack-ansible-deploy-infra_lxc_tls-ubuntu-jammy https://zuul.opendev.org/t/openstack/build/e15128f5dfec471bbc1cc8a30f6cc9e7 : SUCCESS in 1h 19m 28s\n- openstack-ansible-deploy-infra_lxc-ubuntu-noble https://zuul.opendev.org/t/openstack/build/c975364def49487ab9e9c9e8b57bd42a : SUCCESS in 1h 22m 05s\n- openstack-ansible-deploy-infra_lxc_stepca-ubuntu-noble https://zuul.opendev.org/t/openstack/build/dafbd1b92c9d4d35b86f98fd1db727e3 : SUCCESS in 56m 17s\n- openstack-ansible-deploy-infra_lxc_tls-ubuntu-noble https://zuul.opendev.org/t/openstack/build/27210df0d0df43b9a1be2044892e5ce9 : SUCCESS in 47m 49s\n- openstack-ansible-upgrade-infra_lxc-rockylinux-9 https://zuul.opendev.org/t/openstack/build/a07c42f4ee0943009b0405ccfc19971f : SUCCESS in 1h 40m 49s\n- openstack-ansible-upgrade_2024.1-infra_lxc-rockylinux-9 https://zuul.opendev.org/t/openstack/build/45451bc2a73e45ffabb9b26566b1804c : SUCCESS in 1h 41m 46s\n- openstack-ansible-upgrade-infra_lxc-ubuntu-jammy https://zuul.opendev.org/t/openstack/build/16fe77a84cba44f2a9a8d5f38943adc5 : SUCCESS in 1h 28m 57s\n- openstack-ansible-upgrade_2024.1-infra_lxc-ubuntu-jammy https://zuul.opendev.org/t/openstack/build/1b1ca9c82f1c4204973d2f488c2dade0 : SUCCESS in 1h 26m 55s\n- openstack-ansible-deploy-hosts_lxc_zfs-ubuntu-jammy https://zuul.opendev.org/t/openstack/build/9ac61411aba749d1a395e4a666eea019 : SUCCESS in 18m 38s\n- openstack-ansible-deploy-hosts_lxc_btrfs-ubuntu-jammy https://zuul.opendev.org/t/openstack/build/7df5fb05692343c28543c39bc81c60cd : SUCCESS in 21m 11s\n- openstack-ansible-deploy-hosts_lxc_overlayfs-ubuntu-jammy https://zuul.opendev.org/t/openstack/build/2b267816b21040cd8db42bc4d156e20c : SUCCESS in 22m 26s\n- openstack-ansible-deploy-hosts_lxc_zfs-ubuntu-noble https://zuul.opendev.org/t/openstack/build/f3e3be759c9c47e4bb5d2f356a6096ef : SUCCESS in 21m 04s\n- openstack-ansible-deploy-hosts_lxc_btrfs-ubuntu-noble https://zuul.opendev.org/t/openstack/build/a2e906ceb20a4d3cbad766838c210ac4 : SUCCESS in 20m 40s\n- openstack-ansible-deploy-hosts_lxc_overlayfs-ubuntu-noble https://zuul.opendev.org/t/openstack/build/ebe22ab3e0314eb5936864226aceb3a4 : SUCCESS in 19m 59s\n- openstack-ansible-deploy-hosts_lxc_btrfs-debian-bookworm https://zuul.opendev.org/t/openstack/build/2f8107fa3b244033b247683aa90acbe9 : SUCCESS in 19m 26s\n- openstack-ansible-deploy-hosts_lxc_overlayfs-debian-bookworm https://zuul.opendev.org/t/openstack/build/fe2a7eade55b42c09a47b3d4ffd1552a : SUCCESS in 21m 34s\n- openstack-ansible-deploy-hosts_lxc_overlayfs-rockylinux-9 https://zuul.opendev.org/t/openstack/build/1947639841b94acf85c2e8527d9cfed9 : SUCCESS in 31m 39s","accounts_in_message":[],"_revision_number":1},{"id":"f5c60992c950aca3faf7a3cb7ad8172e3f2688cf","author":{"_account_id":28619,"name":"Dmitriy Rabotyagov","email":"noonedeadpunk@gmail.com","username":"noonedeadpunk"},"date":"2025-02-18 12:00:36.000000000","message":"Patch Set 1:\n\n(1 comment)","accounts_in_message":[],"_revision_number":1},{"id":"4a0e6d47fc107081959d0e25834b6e2da7f202d5","tag":"autogenerated:zuul:check","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2025-02-18 14:18:42.000000000","message":"Patch Set 1: Verified+1\n\nBuild succeeded (check pipeline).\nhttps://zuul.opendev.org/t/openstack/buildset/356dde8e20ed46919582ff0af535e8d9\n\n- openstack-ansible-linters-ubuntu-noble https://zuul.opendev.org/t/openstack/build/9f38bca848a74f0c8a4ff61a0741b106 : SUCCESS in 6m 24s\n- openstack-tox-docs https://zuul.opendev.org/t/openstack/build/41f580d7441645348829a0ca1def1690 : SUCCESS in 4m 01s\n- openstack-ansible-deploy-infra_lxc-centos-9-stream https://zuul.opendev.org/t/openstack/build/c9699121acf74b27bb0e9eec0c003728 : SUCCESS in 1h 22m 20s (non-voting)\n- openstack-ansible-deploy-infra_lxc-rockylinux-9 https://zuul.opendev.org/t/openstack/build/29725bc7f7e74169ac6138612e627f2e : SUCCESS in 1h 27m 14s\n- openstack-ansible-deploy-infra_lxc_tls-rockylinux-9 https://zuul.opendev.org/t/openstack/build/ce8a8038041a45f68597551a3dae6a91 : SUCCESS in 1h 31m 48s\n- openstack-ansible-deploy-infra_lxc-debian-bookworm https://zuul.opendev.org/t/openstack/build/1dc39988a86b4e4fb1c6fbe42250a06b : SUCCESS in 1h 11m 46s\n- openstack-ansible-deploy-infra_lxc-ubuntu-jammy https://zuul.opendev.org/t/openstack/build/9d5c168033834ea0b5b8d03dedf705b7 : SUCCESS in 1h 11m 44s\n- openstack-ansible-deploy-infra_lxc_stepca-ubuntu-jammy https://zuul.opendev.org/t/openstack/build/138f9693e9cd4b6494fc78cbee2aef91 : SUCCESS in 1h 15m 35s\n- openstack-ansible-deploy-infra_lxc_tls-ubuntu-jammy https://zuul.opendev.org/t/openstack/build/4c272249b4a049fb89743dcef6207b91 : SUCCESS in 1h 16m 42s\n- openstack-ansible-deploy-infra_lxc-ubuntu-noble https://zuul.opendev.org/t/openstack/build/8d38f07ceed642b88b96135388a1b890 : SUCCESS in 46m 22s\n- openstack-ansible-deploy-infra_lxc_stepca-ubuntu-noble https://zuul.opendev.org/t/openstack/build/2e2c738f9ba149e9be0d4b73fd46266c : SUCCESS in 1h 34m 26s\n- openstack-ansible-deploy-infra_lxc_tls-ubuntu-noble https://zuul.opendev.org/t/openstack/build/88f78d353b0d4ce0a3fc9e7a67d1604b : SUCCESS in 1h 16m 38s\n- openstack-ansible-upgrade-infra_lxc-rockylinux-9 https://zuul.opendev.org/t/openstack/build/ccc5eb98d40d4b4588e147bfd8aebf62 : SUCCESS in 1h 40m 46s\n- openstack-ansible-upgrade_2024.1-infra_lxc-rockylinux-9 https://zuul.opendev.org/t/openstack/build/0405766c55a84c7984599d1aed3961ca : SUCCESS in 2h 10m 48s\n- openstack-ansible-upgrade-infra_lxc-ubuntu-jammy https://zuul.opendev.org/t/openstack/build/df321df495a14b14839f460d1d048ab4 : SUCCESS in 56m 24s\n- openstack-ansible-upgrade_2024.1-infra_lxc-ubuntu-jammy https://zuul.opendev.org/t/openstack/build/653fae5aa4c644799415c8da4f3b74fe : SUCCESS in 1h 40m 50s\n- openstack-ansible-deploy-hosts_lxc_zfs-ubuntu-jammy https://zuul.opendev.org/t/openstack/build/914d8b3d0ff345a2869f85ad475edea7 : SUCCESS in 20m 47s\n- openstack-ansible-deploy-hosts_lxc_btrfs-ubuntu-jammy https://zuul.opendev.org/t/openstack/build/b5bdaf1b709f460991b9d30d23cd85f1 : SUCCESS in 26m 07s\n- openstack-ansible-deploy-hosts_lxc_overlayfs-ubuntu-jammy https://zuul.opendev.org/t/openstack/build/16da5294fe144f0c9e1340ebf0afe8b8 : SUCCESS in 21m 32s\n- openstack-ansible-deploy-hosts_lxc_zfs-ubuntu-noble https://zuul.opendev.org/t/openstack/build/0c2ca9936b7f48b6804d6b918257b27e : SUCCESS in 24m 53s\n- openstack-ansible-deploy-hosts_lxc_btrfs-ubuntu-noble https://zuul.opendev.org/t/openstack/build/9ed7cf3fe84148659390d2c6db99ceea : SUCCESS in 22m 10s\n- openstack-ansible-deploy-hosts_lxc_overlayfs-ubuntu-noble https://zuul.opendev.org/t/openstack/build/3c9c0ad3c6e2497d95f5710007a28370 : SUCCESS in 21m 41s\n- openstack-ansible-deploy-hosts_lxc_btrfs-debian-bookworm https://zuul.opendev.org/t/openstack/build/269a301040ec427b9dbc56a31ff72437 : SUCCESS in 26m 37s\n- openstack-ansible-deploy-hosts_lxc_overlayfs-debian-bookworm https://zuul.opendev.org/t/openstack/build/fcd0280b1dea49bb9e835b7ddbab2279 : SUCCESS in 22m 41s\n- openstack-ansible-deploy-hosts_lxc_overlayfs-rockylinux-9 https://zuul.opendev.org/t/openstack/build/6e3ae6c3558c47eabeec64e0c35d2929 : SUCCESS in 24m 17s","accounts_in_message":[],"_revision_number":1},{"id":"f355174c1c67a9a13e282f9cf14b896511247a83","author":{"_account_id":31542,"name":"Andrew Bonney","email":"andrew.bonney@bbc.co.uk","username":"andrewbonney"},"date":"2025-02-19 08:14:25.000000000","message":"Patch Set 1: Code-Review-1\n\n(2 comments)","accounts_in_message":[],"_revision_number":1},{"id":"2fe067268d037d4097596942c162558f99798664","author":{"_account_id":31542,"name":"Andrew Bonney","email":"andrew.bonney@bbc.co.uk","username":"andrewbonney"},"date":"2025-02-19 09:07:51.000000000","message":"Patch Set 1: Code-Review+2\n\n(3 comments)","accounts_in_message":[],"_revision_number":1},{"id":"74bf76f09f56a389f44f4d922294cf8cd02b03d7","author":{"_account_id":25023,"name":"Jonathan Rosser","email":"jonathan.rosser@rd.bbc.co.uk","username":"jrosser"},"date":"2025-02-19 17:09:53.000000000","message":"Patch Set 1: Code-Review+2 Workflow+1","accounts_in_message":[],"_revision_number":1},{"id":"d913bfb2a43a3153885dd16e1ff5b0fec972c98b","tag":"autogenerated:zuul:gate","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2025-02-19 17:13:13.000000000","message":"Patch Set 1: -Verified\n\nStarting gate jobs.","accounts_in_message":[],"_revision_number":1},{"id":"aadd0baefbc00fe637d763d51214b421feeaff4d","tag":"autogenerated:zuul:gate","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2025-02-19 19:09:38.000000000","message":"Patch Set 1: Verified-2\n\nBuild failed (gate pipeline).  For information on how to proceed, see\nhttps://docs.opendev.org/opendev/infra-manual/latest/developers.html#automated-testing\n\nhttps://zuul.opendev.org/t/openstack/buildset/7801a86f70c442608d7c7c41475b0425\n\n- openstack-ansible-linters-ubuntu-noble https://zuul.opendev.org/t/openstack/build/4b7609b4b8c24549ae6807ffcbe54495 : SUCCESS in 9m 58s\n- openstack-tox-docs https://zuul.opendev.org/t/openstack/build/e43a5266f468423d8f7d81763e0167a4 : SUCCESS in 9m 36s\n- openstack-ansible-deploy-infra_lxc-rockylinux-9 https://zuul.opendev.org/t/openstack/build/54ae319b7e6a4a96a15a4ed13f02b2a9 : SUCCESS in 1h 25m 01s\n- openstack-ansible-deploy-infra_lxc_tls-rockylinux-9 https://zuul.opendev.org/t/openstack/build/609fa09445954ad6997dcd7ce3fe0e8a : SUCCESS in 1h 27m 07s\n- openstack-ansible-deploy-infra_lxc-debian-bookworm https://zuul.opendev.org/t/openstack/build/6a8e91ed6f3443c8a3a14315a9fee48b : SUCCESS in 1h 15m 52s\n- openstack-ansible-deploy-infra_lxc-ubuntu-jammy https://zuul.opendev.org/t/openstack/build/a8e1db5f9aa647a79e8605767ffab541 : SUCCESS in 1h 14m 22s\n- openstack-ansible-deploy-infra_lxc_stepca-ubuntu-jammy https://zuul.opendev.org/t/openstack/build/af9ec67fd3ec4945b353736ed059837e : SUCCESS in 1h 07m 42s\n- openstack-ansible-deploy-infra_lxc_tls-ubuntu-jammy https://zuul.opendev.org/t/openstack/build/161425264a354af0a46bca4908831e06 : SUCCESS in 1h 08m 16s\n- openstack-ansible-deploy-infra_lxc-ubuntu-noble https://zuul.opendev.org/t/openstack/build/31e62fd173994bea840b91b22f91d667 : SUCCESS in 1h 16m 04s\n- openstack-ansible-deploy-infra_lxc_stepca-ubuntu-noble https://zuul.opendev.org/t/openstack/build/894fe7c6d9974a6f9181b5b10ca46da0 : SUCCESS in 52m 45s\n- openstack-ansible-deploy-infra_lxc_tls-ubuntu-noble https://zuul.opendev.org/t/openstack/build/372e70d0d2774a798d56f93441ea82aa : SUCCESS in 1h 14m 30s\n- openstack-ansible-upgrade-infra_lxc-rockylinux-9 https://zuul.opendev.org/t/openstack/build/556476605c524ba2bbdb074f4635e05a : FAILURE in 1h 52m 26s\n- openstack-ansible-upgrade_2024.1-infra_lxc-rockylinux-9 https://zuul.opendev.org/t/openstack/build/9c8ae44c47d54707bf64d95beeab5aad : SUCCESS in 1h 50m 55s\n- openstack-ansible-upgrade-infra_lxc-ubuntu-jammy https://zuul.opendev.org/t/openstack/build/d60c5dddf2554688a35dd21d5e0745ee : SUCCESS in 1h 44m 29s\n- openstack-ansible-upgrade_2024.1-infra_lxc-ubuntu-jammy https://zuul.opendev.org/t/openstack/build/ab2700dc06084829b31dd89dde9c5af7 : SUCCESS in 1h 46m 25s\n- openstack-ansible-deploy-hosts_lxc_zfs-ubuntu-jammy https://zuul.opendev.org/t/openstack/build/83f6e7381c2b4d2bbf2d68d49edb7f76 : SUCCESS in 19m 55s\n- openstack-ansible-deploy-hosts_lxc_btrfs-ubuntu-jammy https://zuul.opendev.org/t/openstack/build/70d6418992c04fd0bda32935fff58418 : SUCCESS in 23m 17s\n- openstack-ansible-deploy-hosts_lxc_overlayfs-ubuntu-jammy https://zuul.opendev.org/t/openstack/build/089d70dbb74941cb9f7a0e3920fb5c0b : SUCCESS in 20m 58s\n- openstack-ansible-deploy-hosts_lxc_zfs-ubuntu-noble https://zuul.opendev.org/t/openstack/build/f3d97e759794465cb767934633367183 : SUCCESS in 25m 47s\n- openstack-ansible-deploy-hosts_lxc_btrfs-ubuntu-noble https://zuul.opendev.org/t/openstack/build/c0d29f1533a040459b3df8c4c428b5a9 : SUCCESS in 22m 12s\n- openstack-ansible-deploy-hosts_lxc_overlayfs-ubuntu-noble https://zuul.opendev.org/t/openstack/build/66064eeefe48479daeb347b2a7e50440 : SUCCESS in 24m 17s\n- openstack-ansible-deploy-hosts_lxc_btrfs-debian-bookworm https://zuul.opendev.org/t/openstack/build/a0b1c5e5c0b5450eb5ae40a3208f67ad : SUCCESS in 21m 33s\n- openstack-ansible-deploy-hosts_lxc_overlayfs-debian-bookworm https://zuul.opendev.org/t/openstack/build/a13a930c4c594deeaa7d1987fcca43d8 : SUCCESS in 22m 18s\n- openstack-ansible-deploy-hosts_lxc_overlayfs-rockylinux-9 https://zuul.opendev.org/t/openstack/build/61f92b3e34834664913997af32a3143b : FAILURE in 24m 06s","accounts_in_message":[],"_revision_number":1},{"id":"bdd750e9d681e598674ab8ad3dbfd8071dcc6bd0","author":{"_account_id":28619,"name":"Dmitriy Rabotyagov","email":"noonedeadpunk@gmail.com","username":"noonedeadpunk"},"date":"2025-02-20 09:06:32.000000000","message":"Patch Set 1:\n\n(1 comment)","accounts_in_message":[],"_revision_number":1},{"id":"56987f08659e056619dcf6b02bded739d0c03b65","tag":"autogenerated:zuul:check","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2025-02-20 11:22:16.000000000","message":"Patch Set 1: Verified-1\n\nBuild failed (check pipeline).  For information on how to proceed, see\nhttps://docs.opendev.org/opendev/infra-manual/latest/developers.html#automated-testing\nand https://docs.openstack.org/project-team-guide/testing.html#how-to-handle-test-failures\n\nhttps://zuul.opendev.org/t/openstack/buildset/339f462f6fb545cf8d038380dde1394b\n\n- openstack-ansible-linters-ubuntu-noble https://zuul.opendev.org/t/openstack/build/bb599f0066704680879fb0ac8d37f3b2 : SUCCESS in 6m 52s\n- openstack-tox-docs https://zuul.opendev.org/t/openstack/build/e409607b8c9b468e86cd57c8a2fbdc4d : SUCCESS in 3m 56s\n- openstack-ansible-deploy-infra_lxc-centos-9-stream https://zuul.opendev.org/t/openstack/build/40e27fc42ed840a68cfd3265f3ea1f14 : SUCCESS in 1h 10m 04s (non-voting)\n- openstack-ansible-deploy-infra_lxc-rockylinux-9 https://zuul.opendev.org/t/openstack/build/a29452db81f3494da102c6fa866ec413 : SUCCESS in 1h 17m 00s\n- openstack-ansible-deploy-infra_lxc_tls-rockylinux-9 https://zuul.opendev.org/t/openstack/build/37c10706f9e74f98ac061a10a9c647cd : SUCCESS in 54m 04s\n- openstack-ansible-deploy-infra_lxc-debian-bookworm https://zuul.opendev.org/t/openstack/build/1b54f63368984b849eaaee474ff45f97 : SUCCESS in 1h 06m 46s\n- openstack-ansible-deploy-infra_lxc-ubuntu-jammy https://zuul.opendev.org/t/openstack/build/30526c6d09f34d5a9a4a66f362d34df1 : SUCCESS in 1h 22m 20s\n- openstack-ansible-deploy-infra_lxc_stepca-ubuntu-jammy https://zuul.opendev.org/t/openstack/build/15678195945c4abdbc2d81b9118c24cb : FAILURE in 28m 05s\n- openstack-ansible-deploy-infra_lxc_tls-ubuntu-jammy https://zuul.opendev.org/t/openstack/build/7cffc86b3f4d436290be64b59159bf4a : SUCCESS in 1h 20m 38s\n- openstack-ansible-deploy-infra_lxc-ubuntu-noble https://zuul.opendev.org/t/openstack/build/d307f71e3e054ec4aabe471b6288ffc1 : SUCCESS in 1h 33m 41s\n- openstack-ansible-deploy-infra_lxc_stepca-ubuntu-noble https://zuul.opendev.org/t/openstack/build/48060cb22ead49449da401db3ec010f3 : SUCCESS in 1h 11m 36s\n- openstack-ansible-deploy-infra_lxc_tls-ubuntu-noble https://zuul.opendev.org/t/openstack/build/134e2cf4a9fb4d6e96612616ebd587d4 : SUCCESS in 1h 14m 43s\n- openstack-ansible-upgrade-infra_lxc-rockylinux-9 https://zuul.opendev.org/t/openstack/build/9400876ab403472383b4eebd4b08ab82 : SUCCESS in 1h 48m 12s\n- openstack-ansible-upgrade_2024.1-infra_lxc-rockylinux-9 https://zuul.opendev.org/t/openstack/build/3f6698dbb9ce4f3fb533346269896b58 : POST_FAILURE in 1h 54m 37s\n- openstack-ansible-upgrade-infra_lxc-ubuntu-jammy https://zuul.opendev.org/t/openstack/build/ff9574df138445fe9b7241cd7aa4f808 : SUCCESS in 1h 29m 36s\n- openstack-ansible-upgrade_2024.1-infra_lxc-ubuntu-jammy https://zuul.opendev.org/t/openstack/build/8cc8cf5119e9484e8b0fd36ee5658de6 : SUCCESS in 1h 34m 36s\n- openstack-ansible-deploy-hosts_lxc_zfs-ubuntu-jammy https://zuul.opendev.org/t/openstack/build/df6f83f3a5a14dcea5377a80be129f2c : FAILURE in 21m 40s\n- openstack-ansible-deploy-hosts_lxc_btrfs-ubuntu-jammy https://zuul.opendev.org/t/openstack/build/2df211c1b6334a4383891b7e99e65075 : SUCCESS in 22m 14s\n- openstack-ansible-deploy-hosts_lxc_overlayfs-ubuntu-jammy https://zuul.opendev.org/t/openstack/build/5404a9f8f23f427286eb3cc661e2fadc : SUCCESS in 21m 20s\n- openstack-ansible-deploy-hosts_lxc_zfs-ubuntu-noble https://zuul.opendev.org/t/openstack/build/5ee1d8b4f7194a51b04f0621361e056f : SUCCESS in 24m 26s\n- openstack-ansible-deploy-hosts_lxc_btrfs-ubuntu-noble https://zuul.opendev.org/t/openstack/build/a143741f365b4554b8eb4a6302f9a1e9 : SUCCESS in 23m 30s\n- openstack-ansible-deploy-hosts_lxc_overlayfs-ubuntu-noble https://zuul.opendev.org/t/openstack/build/314d276e116f4953bfb00f410681b3a2 : SUCCESS in 26m 06s\n- openstack-ansible-deploy-hosts_lxc_btrfs-debian-bookworm https://zuul.opendev.org/t/openstack/build/cfd6a0bd134d4b5ea272dc41320e8994 : SUCCESS in 20m 42s\n- openstack-ansible-deploy-hosts_lxc_overlayfs-debian-bookworm https://zuul.opendev.org/t/openstack/build/be927da31de64db8884caa0d1d1485c1 : SUCCESS in 21m 14s\n- openstack-ansible-deploy-hosts_lxc_overlayfs-rockylinux-9 https://zuul.opendev.org/t/openstack/build/c9a6de7877534e2d9ca30af5814a5ffb : SUCCESS in 25m 12s","accounts_in_message":[],"_revision_number":1},{"id":"9cb0cc912d42811e6f870d4f61ff39d29c8e2968","author":{"_account_id":28619,"name":"Dmitriy Rabotyagov","email":"noonedeadpunk@gmail.com","username":"noonedeadpunk"},"date":"2025-02-20 13:50:17.000000000","message":"Patch Set 1:\n\n(1 comment)","accounts_in_message":[],"_revision_number":1},{"id":"01b902e72bd5d1112c06164d09aeab20b6db4134","tag":"autogenerated:zuul:check","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2025-02-20 15:57:59.000000000","message":"Patch Set 1: Verified+1\n\nBuild succeeded (check pipeline).\nhttps://zuul.opendev.org/t/openstack/buildset/55965c9999634b1ba199d09b1fc6f703\n\n- openstack-ansible-linters-ubuntu-noble https://zuul.opendev.org/t/openstack/build/27000941300e4b9ca0e76551174fbbaa : SUCCESS in 6m 43s\n- openstack-tox-docs https://zuul.opendev.org/t/openstack/build/5fbd8392c17a4af188f65a2640742a2c : SUCCESS in 4m 20s\n- openstack-ansible-deploy-infra_lxc-centos-9-stream https://zuul.opendev.org/t/openstack/build/5fdc8207dd6d4a5186570100605536f3 : SUCCESS in 1h 10m 51s (non-voting)\n- openstack-ansible-deploy-infra_lxc-rockylinux-9 https://zuul.opendev.org/t/openstack/build/93c84705c42240c4882c9432450c9ba7 : SUCCESS in 1h 20m 57s\n- openstack-ansible-deploy-infra_lxc_tls-rockylinux-9 https://zuul.opendev.org/t/openstack/build/464ae4d809dc48928f11dd6d01548423 : SUCCESS in 1h 29m 22s\n- openstack-ansible-deploy-infra_lxc-debian-bookworm https://zuul.opendev.org/t/openstack/build/d776d492226b4b2f8308d690af58bd18 : SUCCESS in 1h 19m 22s\n- openstack-ansible-deploy-infra_lxc-ubuntu-jammy https://zuul.opendev.org/t/openstack/build/72cc01b5b151445bad32ca30acb8e7a2 : SUCCESS in 1h 08m 17s\n- openstack-ansible-deploy-infra_lxc_stepca-ubuntu-jammy https://zuul.opendev.org/t/openstack/build/ba2a478bc2244a90bd8846bcbf548838 : SUCCESS in 1h 24m 36s\n- openstack-ansible-deploy-infra_lxc_tls-ubuntu-jammy https://zuul.opendev.org/t/openstack/build/a3a6863dd22a4e04a8d00bcd02288a36 : SUCCESS in 46m 17s\n- openstack-ansible-deploy-infra_lxc-ubuntu-noble https://zuul.opendev.org/t/openstack/build/ffc54a62708c4e26b2dd28fad08a99d5 : SUCCESS in 1h 16m 42s\n- openstack-ansible-deploy-infra_lxc_stepca-ubuntu-noble https://zuul.opendev.org/t/openstack/build/36a8977ec50b497ba2b7e18a5ccd6c95 : SUCCESS in 2h 05m 12s\n- openstack-ansible-deploy-infra_lxc_tls-ubuntu-noble https://zuul.opendev.org/t/openstack/build/b35de723f4b649f8a85a0a2430b36602 : SUCCESS in 46m 20s\n- openstack-ansible-upgrade-infra_lxc-rockylinux-9 https://zuul.opendev.org/t/openstack/build/44e05baed8844f1bb5ab193c9c7c5a3f : SUCCESS in 1h 55m 50s\n- openstack-ansible-upgrade_2024.1-infra_lxc-rockylinux-9 https://zuul.opendev.org/t/openstack/build/e930408555ae4b7382ef2504efc9d281 : SUCCESS in 1h 58m 57s\n- openstack-ansible-upgrade-infra_lxc-ubuntu-jammy https://zuul.opendev.org/t/openstack/build/3e541e4e3c654f6589ee240b9bab9762 : SUCCESS in 1h 25m 40s\n- openstack-ansible-upgrade_2024.1-infra_lxc-ubuntu-jammy https://zuul.opendev.org/t/openstack/build/9fc396bb406d4593bb850b7c09788322 : SUCCESS in 1h 31m 04s\n- openstack-ansible-deploy-hosts_lxc_zfs-ubuntu-jammy https://zuul.opendev.org/t/openstack/build/fb754833ea094cd4879233c9794f4d2b : SUCCESS in 21m 43s\n- openstack-ansible-deploy-hosts_lxc_btrfs-ubuntu-jammy https://zuul.opendev.org/t/openstack/build/9cb94bdc60cd4618bccb8bdaa84a79de : SUCCESS in 19m 34s\n- openstack-ansible-deploy-hosts_lxc_overlayfs-ubuntu-jammy https://zuul.opendev.org/t/openstack/build/7912ac2705e34e97b3aa54338375dd3e : SUCCESS in 20m 03s\n- openstack-ansible-deploy-hosts_lxc_zfs-ubuntu-noble https://zuul.opendev.org/t/openstack/build/524d88d5de7d4dd281361ad1bb657015 : SUCCESS in 20m 01s\n- openstack-ansible-deploy-hosts_lxc_btrfs-ubuntu-noble https://zuul.opendev.org/t/openstack/build/d9b9a3c4e2a7436fbe6cdbe40c2eecb1 : SUCCESS in 19m 44s\n- openstack-ansible-deploy-hosts_lxc_overlayfs-ubuntu-noble https://zuul.opendev.org/t/openstack/build/92cdaa72feab434ab2cb809168e485af : SUCCESS in 26m 37s\n- openstack-ansible-deploy-hosts_lxc_btrfs-debian-bookworm https://zuul.opendev.org/t/openstack/build/18c3d82b445a44bd8f88940f8ef13f49 : SUCCESS in 28m 00s\n- openstack-ansible-deploy-hosts_lxc_overlayfs-debian-bookworm https://zuul.opendev.org/t/openstack/build/5d9e54558b454e429e1f8cfa4c7659f1 : SUCCESS in 21m 26s\n- openstack-ansible-deploy-hosts_lxc_overlayfs-rockylinux-9 https://zuul.opendev.org/t/openstack/build/9d316bd16b0f42719abbe214c28fb5d3 : SUCCESS in 23m 40s","accounts_in_message":[],"_revision_number":1},{"id":"ec6da7065664a2caeb2c8459e9235bfb4097f817","tag":"autogenerated:zuul:gate","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2025-02-20 16:00:10.000000000","message":"Patch Set 1: -Verified\n\nStarting gate jobs.","accounts_in_message":[],"_revision_number":1},{"id":"f13508deb9f2e59e301fc0b7e8b52fee4c8b21a7","tag":"autogenerated:zuul:gate","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2025-02-20 18:14:04.000000000","message":"Patch Set 1: Verified+2\n\nBuild succeeded (gate pipeline).\nhttps://zuul.opendev.org/t/openstack/buildset/3e2243e770f146a1a8894908ff718415\n\n- openstack-ansible-linters-ubuntu-noble https://zuul.opendev.org/t/openstack/build/1a62c6a6480548c58fccdfec25858088 : SUCCESS in 9m 02s\n- openstack-tox-docs https://zuul.opendev.org/t/openstack/build/e5bd8b153f3e4cda87ce43e0d143b7f6 : SUCCESS in 4m 04s\n- openstack-ansible-deploy-infra_lxc-rockylinux-9 https://zuul.opendev.org/t/openstack/build/5e2265442d1041a5a5861d10c09df963 : SUCCESS in 1h 21m 02s\n- openstack-ansible-deploy-infra_lxc_tls-rockylinux-9 https://zuul.opendev.org/t/openstack/build/0bab82f5935e40f7aa1353c273eb9555 : SUCCESS in 1h 27m 54s\n- openstack-ansible-deploy-infra_lxc-debian-bookworm https://zuul.opendev.org/t/openstack/build/d2fb2782525249c89af2976d4853b9fa : SUCCESS in 1h 07m 24s\n- openstack-ansible-deploy-infra_lxc-ubuntu-jammy https://zuul.opendev.org/t/openstack/build/f12513f0f9cb49ed8439fdb807b8dd9f : SUCCESS in 1h 25m 56s\n- openstack-ansible-deploy-infra_lxc_stepca-ubuntu-jammy https://zuul.opendev.org/t/openstack/build/e29df8b9b95c4373a36dd929197fa25f : SUCCESS in 1h 10m 28s\n- openstack-ansible-deploy-infra_lxc_tls-ubuntu-jammy https://zuul.opendev.org/t/openstack/build/e3d44cddd4b64ff384a4d261e6d95f61 : SUCCESS in 1h 08m 47s\n- openstack-ansible-deploy-infra_lxc-ubuntu-noble https://zuul.opendev.org/t/openstack/build/cbba7404730642e382729fc0e16fdacb : SUCCESS in 1h 08m 37s\n- openstack-ansible-deploy-infra_lxc_stepca-ubuntu-noble https://zuul.opendev.org/t/openstack/build/3a203010f741462f8a7c8178dcd7aa58 : SUCCESS in 1h 25m 23s\n- openstack-ansible-deploy-infra_lxc_tls-ubuntu-noble https://zuul.opendev.org/t/openstack/build/317eff225bee4e05b2e2acb9430cbd9a : SUCCESS in 1h 13m 26s\n- openstack-ansible-upgrade-infra_lxc-rockylinux-9 https://zuul.opendev.org/t/openstack/build/36f61c5025ea424aa53122aae8c0889a : SUCCESS in 1h 45m 59s\n- openstack-ansible-upgrade_2024.1-infra_lxc-rockylinux-9 https://zuul.opendev.org/t/openstack/build/1ae3f5352fdf4a39a0315b4c331f7f66 : SUCCESS in 1h 56m 41s\n- openstack-ansible-upgrade-infra_lxc-ubuntu-jammy https://zuul.opendev.org/t/openstack/build/6d47457f11b343b79507ae6531338727 : SUCCESS in 1h 29m 04s\n- openstack-ansible-upgrade_2024.1-infra_lxc-ubuntu-jammy https://zuul.opendev.org/t/openstack/build/155cb93087814d9591e95db3c2fa782b : SUCCESS in 1h 22m 58s\n- openstack-ansible-deploy-hosts_lxc_zfs-ubuntu-jammy https://zuul.opendev.org/t/openstack/build/893416761661466385b5de8fdb13dfcd : SUCCESS in 21m 54s\n- openstack-ansible-deploy-hosts_lxc_btrfs-ubuntu-jammy https://zuul.opendev.org/t/openstack/build/ca0443af65f341a48303f38c6ec31e94 : SUCCESS in 25m 08s\n- openstack-ansible-deploy-hosts_lxc_overlayfs-ubuntu-jammy https://zuul.opendev.org/t/openstack/build/f8f5788694db43e2bead4e180473fe26 : SUCCESS in 20m 38s\n- openstack-ansible-deploy-hosts_lxc_zfs-ubuntu-noble https://zuul.opendev.org/t/openstack/build/d06b8625eb0d448985e306fc5e062b5c : SUCCESS in 17m 05s\n- openstack-ansible-deploy-hosts_lxc_btrfs-ubuntu-noble https://zuul.opendev.org/t/openstack/build/c17a1f5d9c624d1881cc25d9af704c60 : SUCCESS in 22m 16s\n- openstack-ansible-deploy-hosts_lxc_overlayfs-ubuntu-noble https://zuul.opendev.org/t/openstack/build/9a1bf79b1e1b4198afdfce7f5de91e05 : SUCCESS in 25m 48s\n- openstack-ansible-deploy-hosts_lxc_btrfs-debian-bookworm https://zuul.opendev.org/t/openstack/build/8f1613e0575b4d80b891d02ded5a7b13 : SUCCESS in 21m 26s\n- openstack-ansible-deploy-hosts_lxc_overlayfs-debian-bookworm https://zuul.opendev.org/t/openstack/build/e31b502468a049f59b2d35fc6b8cd5d4 : SUCCESS in 22m 38s\n- openstack-ansible-deploy-hosts_lxc_overlayfs-rockylinux-9 https://zuul.opendev.org/t/openstack/build/26843c24fa6e4ea7b19c55c2ad9577fb : SUCCESS in 23m 52s","accounts_in_message":[],"_revision_number":1},{"id":"806a846a6034c8d566f931fb38dcf5845b85c826","tag":"autogenerated:gerrit:merged","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2025-02-20 18:14:04.000000000","message":"Change has been successfully merged","accounts_in_message":[],"_revision_number":1},{"id":"bea23d6c765a91442506936966b24d2315944cf3","tag":"autogenerated:zuul:promote","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2025-02-20 18:15:06.000000000","message":"Patch Set 1:\n\nBuild succeeded (promote pipeline).\nhttps://zuul.opendev.org/t/openstack/buildset/8072bde5472349658340a598ef688500\n\n- promote-openstack-tox-docs https://zuul.opendev.org/t/openstack/build/bdb6e8adce5e4ebb975cda7c8df2d883 : SUCCESS in 38s","accounts_in_message":[],"_revision_number":1}],"current_revision_number":1,"current_revision":"2aed220123e994df349251eb63300ce9e78c1475","revisions":{"2aed220123e994df349251eb63300ce9e78c1475":{"kind":"REWORK","_number":1,"created":"2025-02-13 08:36:25.000000000","uploader":{"_account_id":28619,"name":"Dmitriy Rabotyagov","email":"noonedeadpunk@gmail.com","username":"noonedeadpunk"},"ref":"refs/changes/28/941528/1","fetch":{"anonymous http":{"url":"https://review.opendev.org/openstack/openstack-ansible-lxc_hosts","ref":"refs/changes/28/941528/1","commands":{"Checkout":"git fetch https://review.opendev.org/openstack/openstack-ansible-lxc_hosts refs/changes/28/941528/1 \u0026\u0026 git checkout FETCH_HEAD","Cherry Pick":"git fetch https://review.opendev.org/openstack/openstack-ansible-lxc_hosts refs/changes/28/941528/1 \u0026\u0026 git cherry-pick FETCH_HEAD","Format Patch":"git fetch https://review.opendev.org/openstack/openstack-ansible-lxc_hosts refs/changes/28/941528/1 \u0026\u0026 git format-patch -1 --stdout FETCH_HEAD","Pull":"git pull https://review.opendev.org/openstack/openstack-ansible-lxc_hosts refs/changes/28/941528/1"}}},"commit":{"parents":[{"commit":"22b605abe401e18dfd7990d6fab10fb7b98ed7fb","subject":"Make sysctl configuration path configurable Defaults to /etc/sysctl.conf to retain current behavior","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/openstack-ansible-lxc_hosts/commit/22b605abe401e18dfd7990d6fab10fb7b98ed7fb"}]}],"author":{"name":"Dmitriy Rabotyagov","email":"dmitriy.rabotyagov@cleura.com","date":"2025-02-13 08:36:17.000000000","tz":60},"committer":{"name":"Dmitriy Rabotyagov","email":"dmitriy.rabotyagov@cleura.com","date":"2025-02-13 08:36:17.000000000","tz":60},"subject":"Auto-fix yaml rules","message":"Auto-fix yaml rules\n\nIn order to reduce divergance with ansible-lint rules, we apply\nauto-fixing of violations.\n\nIn current patch we replace all kind of truthy variables with\n`true` or `false` values to align with recommendations along with\nalignment of used quotes.\n\nChange-Id: I84f2efb55945e0379f44fca8ddb7b92241c1d9ef\n","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/openstack-ansible-lxc_hosts/commit/2aed220123e994df349251eb63300ce9e78c1475"}],"resolve_conflicts_web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/openstack-ansible-lxc_hosts/commit/2aed220123e994df349251eb63300ce9e78c1475"}]},"branch":"refs/heads/master"}},"requirements":[],"submit_records":[{"rule_name":"gerrit~DefaultSubmitRule","status":"CLOSED","labels":[{"label":"Verified","status":"MAY","applied_by":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]}},{"label":"Code-Review","status":"MAY","applied_by":{"_account_id":31542,"name":"Andrew Bonney","email":"andrew.bonney@bbc.co.uk","username":"andrewbonney"}},{"label":"Workflow","status":"MAY","applied_by":{"_account_id":25023,"name":"Jonathan Rosser","email":"jonathan.rosser@rd.bbc.co.uk","username":"jrosser"}},{"label":"Backport-Candidate","status":"MAY"}]}],"submit_requirements":[{"name":"Verified","description":"Verified in gate by CI","status":"SATISFIED","is_legacy":false,"submittability_expression_result":{"expression":"label:Verified\u003dMAX AND -label:Verified\u003dMIN","fulfilled":true,"status":"PASS","passing_atoms":["label:Verified\u003dMAX"],"failing_atoms":["label:Verified\u003dMIN"],"atom_explanations":{}}},{"name":"Backport-Candidate","description":"Backport candidate status","status":"NOT_APPLICABLE","is_legacy":false,"applicability_expression_result":{"fulfilled":false,"status":"FAIL"},"submittability_expression_result":{"expression":"is:true","fulfilled":true,"status":"NOT_EVALUATED","passing_atoms":[],"failing_atoms":[],"atom_explanations":{}}},{"name":"Code-Review","description":"Code reviewed by core reviewer","status":"SATISFIED","is_legacy":false,"submittability_expression_result":{"expression":"label:Code-Review\u003dMAX AND -label:Code-Review\u003dMIN","fulfilled":true,"status":"PASS","passing_atoms":["label:Code-Review\u003dMAX"],"failing_atoms":["label:Code-Review\u003dMIN"],"atom_explanations":{}}},{"name":"Workflow","description":"Approved for gate by core reviewer","status":"SATISFIED","is_legacy":false,"submittability_expression_result":{"expression":"label:Workflow\u003dMAX AND -label:Workflow\u003dMIN","fulfilled":true,"status":"PASS","passing_atoms":["label:Workflow\u003dMAX"],"failing_atoms":["label:Workflow\u003dMIN"],"atom_explanations":{}}}]}
