)]}'
{"id":"openstack%2Fopenstack-ansible-lxc_hosts~945229","triplet_id":"openstack%2Fopenstack-ansible-lxc_hosts~stable%2F2024.2~Ie5b0f215cceb8e8002b75543f4050042732525a7","project":"openstack/openstack-ansible-lxc_hosts","branch":"stable/2024.2","topic":"osa/fqcn-stable/2024.2","attention_set":{},"removed_from_attention_set":{"28619":{"account":{"_account_id":28619,"name":"Dmitriy Rabotyagov","email":"noonedeadpunk@gmail.com","username":"noonedeadpunk"},"last_update":"2025-03-21 21:00:13.000000000","reason":"Change was submitted"}},"hashtags":[],"change_id":"Ie5b0f215cceb8e8002b75543f4050042732525a7","subject":"Use distro-specific SSH service name","status":"MERGED","created":"2025-03-21 15:29:53.000000000","updated":"2025-03-21 21:01:06.000000000","submitted":"2025-03-21 21:00:13.000000000","submitter":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"total_comment_count":0,"unresolved_comment_count":0,"has_review_started":true,"submission_id":"945229-osa/fqcn-stable/2024.2","cherry_pick_of_change":942812,"cherry_pick_of_patch_set":1,"meta_rev_id":"3bb970ae961164f102f61f5e6f4914d7a35b1060","_number":945229,"virtual_id_number":945229,"owner":{"_account_id":28619,"name":"Dmitriy Rabotyagov","email":"noonedeadpunk@gmail.com","username":"noonedeadpunk"},"actions":{},"labels":{"Verified":{"approved":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"all":[{"value":0,"_account_id":28619,"name":"Dmitriy Rabotyagov","email":"noonedeadpunk@gmail.com","username":"noonedeadpunk"},{"tag":"autogenerated:zuul:gate","value":2,"date":"2025-03-21 21:00:13.000000000","permitted_voting_range":{"min":2,"max":2},"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},{"value":0,"_account_id":25023,"name":"Jonathan Rosser","email":"jonathan.rosser@rd.bbc.co.uk","username":"jrosser"}],"values":{"-2":"Fails","-1":"Doesn\u0027t seem to work"," 0":"No score","+1":"Works for me","+2":"Verified"},"description":"","default_value":0,"optional":true},"Code-Review":{"approved":{"_account_id":28619,"name":"Dmitriy Rabotyagov","email":"noonedeadpunk@gmail.com","username":"noonedeadpunk"},"all":[{"value":2,"date":"2025-03-21 15:31:36.000000000","permitted_voting_range":{"min":2,"max":2},"_account_id":28619,"name":"Dmitriy Rabotyagov","email":"noonedeadpunk@gmail.com","username":"noonedeadpunk"},{"value":0,"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},{"value":2,"date":"2025-03-21 18:59:53.000000000","permitted_voting_range":{"min":2,"max":2},"_account_id":25023,"name":"Jonathan Rosser","email":"jonathan.rosser@rd.bbc.co.uk","username":"jrosser"}],"values":{"-2":"Do not merge","-1":"This patch needs further work before it can be merged"," 0":"No score","+1":"Looks good to me, but someone else must approve","+2":"Looks good to me (core reviewer)"},"description":"","default_value":0,"optional":true},"Workflow":{"approved":{"_account_id":25023,"name":"Jonathan Rosser","email":"jonathan.rosser@rd.bbc.co.uk","username":"jrosser"},"all":[{"value":0,"_account_id":28619,"name":"Dmitriy Rabotyagov","email":"noonedeadpunk@gmail.com","username":"noonedeadpunk"},{"value":0,"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},{"value":1,"date":"2025-03-21 18:59:53.000000000","permitted_voting_range":{"min":1,"max":1},"_account_id":25023,"name":"Jonathan Rosser","email":"jonathan.rosser@rd.bbc.co.uk","username":"jrosser"}],"values":{"-1":"Work in progress"," 0":"Ready for reviews","+1":"Approved"},"description":"","default_value":0,"optional":true},"Backport-Candidate":{"all":[{"value":0,"_account_id":28619,"name":"Dmitriy Rabotyagov","email":"noonedeadpunk@gmail.com","username":"noonedeadpunk"},{"value":0,"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},{"value":0,"_account_id":25023,"name":"Jonathan Rosser","email":"jonathan.rosser@rd.bbc.co.uk","username":"jrosser"}],"values":{"-1":"Not A Backport Candidate"," 0":"Backport Review Needed","+1":"Proposed Backport"},"description":"","default_value":0,"optional":true}},"removable_reviewers":[],"reviewers":{"REVIEWER":[{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},{"_account_id":25023,"name":"Jonathan Rosser","email":"jonathan.rosser@rd.bbc.co.uk","username":"jrosser"},{"_account_id":28619,"name":"Dmitriy Rabotyagov","email":"noonedeadpunk@gmail.com","username":"noonedeadpunk"}]},"pending_reviewers":{},"reviewer_updates":[{"updated":"2025-03-21 17:36:10.000000000","updated_by":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"reviewer":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"state":"REVIEWER"},{"updated":"2025-03-21 18:59:53.000000000","updated_by":{"_account_id":25023,"name":"Jonathan Rosser","email":"jonathan.rosser@rd.bbc.co.uk","username":"jrosser"},"reviewer":{"_account_id":25023,"name":"Jonathan Rosser","email":"jonathan.rosser@rd.bbc.co.uk","username":"jrosser"},"state":"REVIEWER"}],"messages":[{"id":"c19cda035f982fa54b173594271284c03ba9c6f2","tag":"autogenerated:gerrit:newWipPatchSet","author":{"_account_id":28619,"name":"Dmitriy Rabotyagov","email":"noonedeadpunk@gmail.com","username":"noonedeadpunk"},"date":"2025-03-21 15:29:53.000000000","message":"Patch Set 1: Cherry Picked from branch master.\n\nThe following files contain Git conflicts:\n* handlers/main.yml","accounts_in_message":[],"_revision_number":1},{"id":"929bc96ba6bc7e91debe5693abf525d4f6a1e908","tag":"autogenerated:gerrit:newWipPatchSet","author":{"_account_id":28619,"name":"Dmitriy Rabotyagov","email":"noonedeadpunk@gmail.com","username":"noonedeadpunk"},"date":"2025-03-21 15:30:45.000000000","message":"Patch Set 2: Published edit on patch set 1.","accounts_in_message":[],"_revision_number":2},{"id":"2cbce00d686603ed90436968834a0cb9e73e239a","tag":"autogenerated:gerrit:setReadyForReview","author":{"_account_id":28619,"name":"Dmitriy Rabotyagov","email":"noonedeadpunk@gmail.com","username":"noonedeadpunk"},"date":"2025-03-21 15:30:48.000000000","message":"Set Ready For Review","accounts_in_message":[],"_revision_number":2},{"id":"aa2171720cff4d8890f73e25b3ab36059cf03524","tag":"autogenerated:gerrit:newPatchSet","author":{"_account_id":28619,"name":"Dmitriy Rabotyagov","email":"noonedeadpunk@gmail.com","username":"noonedeadpunk"},"date":"2025-03-21 15:31:32.000000000","message":"Patch Set 3: Published edit on patch set 2.","accounts_in_message":[],"_revision_number":3},{"id":"94538af3b64b62aa90c97732f9515367c3850c88","author":{"_account_id":28619,"name":"Dmitriy Rabotyagov","email":"noonedeadpunk@gmail.com","username":"noonedeadpunk"},"date":"2025-03-21 15:31:36.000000000","message":"Patch Set 3: Code-Review+2","accounts_in_message":[],"_revision_number":3},{"id":"c0b3a5d332189d1345ab65417a7aedbeb68740c4","tag":"autogenerated:zuul:check","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2025-03-21 17:36:10.000000000","message":"Patch Set 3: Verified+1\n\nBuild succeeded (check pipeline).\nhttps://zuul.opendev.org/t/openstack/buildset/5eefbb1525054186b5e10600596af308\n\n- openstack-ansible-linters-ubuntu-noble https://zuul.opendev.org/t/openstack/build/03388cb773f245f98e68255a11619ec2 : SUCCESS in 8m 58s\n- openstack-tox-docs https://zuul.opendev.org/t/openstack/build/71117a508e8a48f99f71edf33f7ef190 : SUCCESS in 4m 40s\n- openstack-ansible-deploy-infra_lxc-centos-9-stream https://zuul.opendev.org/t/openstack/build/4bc8d5ba973d43c6834b2627b1d402c8 : SUCCESS in 1h 19m 07s (non-voting)\n- openstack-ansible-deploy-infra_lxc-rockylinux-9 https://zuul.opendev.org/t/openstack/build/68138dea42604ce6a0cd7ac06e063674 : SUCCESS in 1h 22m 20s\n- openstack-ansible-deploy-infra_lxc_tls-rockylinux-9 https://zuul.opendev.org/t/openstack/build/fbddf16b2ea04c5994efd2bec27904c9 : SUCCESS in 48m 30s\n- openstack-ansible-deploy-infra_lxc-debian-bookworm https://zuul.opendev.org/t/openstack/build/ebf80f4555dd4a36ace4118572d0b6da : SUCCESS in 1h 19m 57s\n- openstack-ansible-deploy-infra_lxc-ubuntu-jammy https://zuul.opendev.org/t/openstack/build/99453d5e600f4962a4388aa0a2d619ae : SUCCESS in 1h 04m 00s\n- openstack-ansible-deploy-infra_lxc_stepca-ubuntu-jammy https://zuul.opendev.org/t/openstack/build/da008428fcb44bf3aadebfc9cf747bfa : SUCCESS in 1h 08m 20s\n- openstack-ansible-deploy-infra_lxc_tls-ubuntu-jammy https://zuul.opendev.org/t/openstack/build/11cf2935c94348798e60b9f9d68d8efd : SUCCESS in 38m 44s\n- openstack-ansible-deploy-infra_lxc-ubuntu-noble https://zuul.opendev.org/t/openstack/build/85f73775db784274a9931c3be9f8d17f : SUCCESS in 44m 52s\n- openstack-ansible-deploy-infra_lxc_stepca-ubuntu-noble https://zuul.opendev.org/t/openstack/build/0ded63a7a3874627b3c0e600224ba099 : SUCCESS in 1h 23m 19s\n- openstack-ansible-deploy-infra_lxc_tls-ubuntu-noble https://zuul.opendev.org/t/openstack/build/eed01fb45282421bb1d56f4ba1e95717 : SUCCESS in 1h 05m 59s\n- openstack-ansible-upgrade-infra_lxc-rockylinux-9 https://zuul.opendev.org/t/openstack/build/5f24baf238ef42b9a7a30e021c68c120 : SUCCESS in 2h 02m 24s\n- openstack-ansible-upgrade-infra_lxc-ubuntu-jammy https://zuul.opendev.org/t/openstack/build/3a67eae15998433599d35dba8354170f : SUCCESS in 1h 25m 24s\n- openstack-ansible-deploy-hosts_lxc_zfs-ubuntu-jammy https://zuul.opendev.org/t/openstack/build/7b08dc61678342998b7f54335439a3fe : SUCCESS in 21m 05s\n- openstack-ansible-deploy-hosts_lxc_btrfs-ubuntu-jammy https://zuul.opendev.org/t/openstack/build/ec947316f7a34b4b9c0c0cd84cf04712 : SUCCESS in 12m 27s\n- openstack-ansible-deploy-hosts_lxc_overlayfs-ubuntu-jammy https://zuul.opendev.org/t/openstack/build/fa4b1a70fe324973a734083225b8c2ee : SUCCESS in 14m 23s\n- openstack-ansible-deploy-hosts_lxc_zfs-ubuntu-noble https://zuul.opendev.org/t/openstack/build/ec0ab252a64d474dba6199fb4958f442 : SUCCESS in 21m 20s\n- openstack-ansible-deploy-hosts_lxc_btrfs-ubuntu-noble https://zuul.opendev.org/t/openstack/build/c1a64a96cd984f6894985a9a34b721d9 : SUCCESS in 21m 11s\n- openstack-ansible-deploy-hosts_lxc_overlayfs-ubuntu-noble https://zuul.opendev.org/t/openstack/build/d02e032f7536417da15e32037abd45e6 : SUCCESS in 24m 54s\n- openstack-ansible-deploy-hosts_lxc_btrfs-debian-bookworm https://zuul.opendev.org/t/openstack/build/54bfa455b5a44bc397a9118bbc3bc11f : SUCCESS in 22m 00s\n- openstack-ansible-deploy-hosts_lxc_overlayfs-debian-bookworm https://zuul.opendev.org/t/openstack/build/a13f75078a424fc2ba1260cf559f6e4f : SUCCESS in 22m 20s\n- openstack-ansible-deploy-hosts_lxc_overlayfs-rockylinux-9 https://zuul.opendev.org/t/openstack/build/5429ae9607db4d8b9273e70bce77f2b0 : SUCCESS in 24m 33s","accounts_in_message":[],"_revision_number":3},{"id":"53e0e29fd3332217c335d4a3b75ad99d2d7bf272","author":{"_account_id":25023,"name":"Jonathan Rosser","email":"jonathan.rosser@rd.bbc.co.uk","username":"jrosser"},"date":"2025-03-21 18:59:53.000000000","message":"Patch Set 3: Code-Review+2 Workflow+1","accounts_in_message":[],"_revision_number":3},{"id":"51e02d2b4bbf95724bc8276f5be8638bc969fc3d","tag":"autogenerated:zuul:gate","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2025-03-21 19:02:02.000000000","message":"Patch Set 3: -Verified\n\nStarting gate jobs.","accounts_in_message":[],"_revision_number":3},{"id":"423675bad75f6f03ccab5b603dcb02d185941738","tag":"autogenerated:zuul:gate","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2025-03-21 21:00:13.000000000","message":"Patch Set 3: Verified+2\n\nBuild succeeded (gate pipeline).\nhttps://zuul.opendev.org/t/openstack/buildset/07e176c4dfd641268cf445b83abdd942\n\n- openstack-ansible-linters-ubuntu-noble https://zuul.opendev.org/t/openstack/build/73e776a7d35a43dda18c10b7c77b4abb : SUCCESS in 6m 44s\n- openstack-tox-docs https://zuul.opendev.org/t/openstack/build/426f9bdaf061440ab664ca8ee787cd87 : SUCCESS in 3m 33s\n- openstack-ansible-deploy-infra_lxc-rockylinux-9 https://zuul.opendev.org/t/openstack/build/c0a36eaf8e124d0e956997711aafebe3 : SUCCESS in 1h 08m 38s\n- openstack-ansible-deploy-infra_lxc_tls-rockylinux-9 https://zuul.opendev.org/t/openstack/build/cbca34cbb9a445299fa90e1ba482e6d6 : SUCCESS in 47m 12s\n- openstack-ansible-deploy-infra_lxc-debian-bookworm https://zuul.opendev.org/t/openstack/build/3addd10a37d9412b921f80e1bd92e6f9 : SUCCESS in 42m 32s\n- openstack-ansible-deploy-infra_lxc-ubuntu-jammy https://zuul.opendev.org/t/openstack/build/99fb6bac255344a7b8ad442f1b4f99be : SUCCESS in 39m 23s\n- openstack-ansible-deploy-infra_lxc_stepca-ubuntu-jammy https://zuul.opendev.org/t/openstack/build/57d58040e57349baab8bf6dd1ff69f68 : SUCCESS in 1h 12m 11s\n- openstack-ansible-deploy-infra_lxc_tls-ubuntu-jammy https://zuul.opendev.org/t/openstack/build/59bd638a6689440f8fe407fac15081dd : SUCCESS in 57m 36s\n- openstack-ansible-deploy-infra_lxc-ubuntu-noble https://zuul.opendev.org/t/openstack/build/e73ce66184e84cec9e403c84852ab502 : SUCCESS in 39m 01s\n- openstack-ansible-deploy-infra_lxc_stepca-ubuntu-noble https://zuul.opendev.org/t/openstack/build/5cfeb0a37e4445c6bcf37a05db11f543 : SUCCESS in 1h 16m 43s\n- openstack-ansible-deploy-infra_lxc_tls-ubuntu-noble https://zuul.opendev.org/t/openstack/build/3c0ce35f5181409bb26ee8cf8ada765e : SUCCESS in 42m 02s\n- openstack-ansible-upgrade-infra_lxc-rockylinux-9 https://zuul.opendev.org/t/openstack/build/79ac77dd98cc4412999f632fa15b0cd2 : SUCCESS in 1h 58m 04s\n- openstack-ansible-upgrade-infra_lxc-ubuntu-jammy https://zuul.opendev.org/t/openstack/build/964e77afd2b84b93bb5ba7de149ce456 : SUCCESS in 1h 46m 36s\n- openstack-ansible-deploy-hosts_lxc_zfs-ubuntu-jammy https://zuul.opendev.org/t/openstack/build/a0867c148e6a4f5b9cb35032f819006b : SUCCESS in 24m 03s\n- openstack-ansible-deploy-hosts_lxc_btrfs-ubuntu-jammy https://zuul.opendev.org/t/openstack/build/bb15166f4fc5409d83e1e8543e4c735c : SUCCESS in 20m 48s\n- openstack-ansible-deploy-hosts_lxc_overlayfs-ubuntu-jammy https://zuul.opendev.org/t/openstack/build/2d2ec34beb1344b2ad4776cb3d9a7ff7 : SUCCESS in 19m 30s\n- openstack-ansible-deploy-hosts_lxc_zfs-ubuntu-noble https://zuul.opendev.org/t/openstack/build/cdb3400dbbe84dce943266d77d03c21a : SUCCESS in 24m 52s\n- openstack-ansible-deploy-hosts_lxc_btrfs-ubuntu-noble https://zuul.opendev.org/t/openstack/build/68d13fb115b74d51be8e35352ee43a8c : SUCCESS in 22m 21s\n- openstack-ansible-deploy-hosts_lxc_overlayfs-ubuntu-noble https://zuul.opendev.org/t/openstack/build/a4ef626f6ef04e339be42e7f38aa1b51 : SUCCESS in 26m 44s\n- openstack-ansible-deploy-hosts_lxc_btrfs-debian-bookworm https://zuul.opendev.org/t/openstack/build/f89af127a5a44bd2bc0a96949abe099b : SUCCESS in 20m 30s\n- openstack-ansible-deploy-hosts_lxc_overlayfs-debian-bookworm https://zuul.opendev.org/t/openstack/build/e2aeb19a40b145ebb7c118fa94b4f2e2 : SUCCESS in 20m 59s\n- openstack-ansible-deploy-hosts_lxc_overlayfs-rockylinux-9 https://zuul.opendev.org/t/openstack/build/1daa084f6d264cbea8146b444bc8ab85 : SUCCESS in 21m 44s","accounts_in_message":[],"_revision_number":3},{"id":"ebe635721a5e3ed269dcbbe011ac15f588f8d1b9","tag":"autogenerated:gerrit:merged","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2025-03-21 21:00:13.000000000","message":"Change has been successfully merged","accounts_in_message":[],"_revision_number":3},{"id":"3bb970ae961164f102f61f5e6f4914d7a35b1060","tag":"autogenerated:zuul:promote","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2025-03-21 21:01:06.000000000","message":"Patch Set 3:\n\nBuild succeeded (promote pipeline).\nhttps://zuul.opendev.org/t/openstack/buildset/89f532b32bf142ca9a861241a4ca79ad\n\n- promote-openstack-tox-docs https://zuul.opendev.org/t/openstack/build/abe26a34027546888e2064d035fbed77 : SUCCESS in 36s","accounts_in_message":[],"_revision_number":3}],"current_revision_number":3,"current_revision":"504024d047e066d767e9b71c10931d0c8c8d42ba","revisions":{"3739ef843f99c40249ea859615fab1a33ae4bd70":{"kind":"REWORK","_number":1,"created":"2025-03-21 15:29:53.000000000","uploader":{"_account_id":28619,"name":"Dmitriy Rabotyagov","email":"noonedeadpunk@gmail.com","username":"noonedeadpunk"},"ref":"refs/changes/29/945229/1","fetch":{"anonymous http":{"url":"https://review.opendev.org/openstack/openstack-ansible-lxc_hosts","ref":"refs/changes/29/945229/1","commands":{"Checkout":"git fetch https://review.opendev.org/openstack/openstack-ansible-lxc_hosts refs/changes/29/945229/1 \u0026\u0026 git checkout FETCH_HEAD","Cherry Pick":"git fetch https://review.opendev.org/openstack/openstack-ansible-lxc_hosts refs/changes/29/945229/1 \u0026\u0026 git cherry-pick FETCH_HEAD","Format Patch":"git fetch https://review.opendev.org/openstack/openstack-ansible-lxc_hosts refs/changes/29/945229/1 \u0026\u0026 git format-patch -1 --stdout FETCH_HEAD","Pull":"git pull https://review.opendev.org/openstack/openstack-ansible-lxc_hosts refs/changes/29/945229/1"}}},"commit":{"parents":[{"commit":"41470784fd9dc2438799f45417a9fbcef0d069d3","subject":"Update TOX_CONSTRAINTS_FILE for stable/2024.2","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/openstack-ansible-lxc_hosts/commit/41470784fd9dc2438799f45417a9fbcef0d069d3"}]}],"author":{"name":"Dmitriy Rabotyagov","email":"dmitriy.rabotyagov@cleura.com","date":"2025-02-26 14:33:52.000000000","tz":60},"committer":{"name":"Dmitriy Rabotyagov","email":"dmitriy.rabotyagov@cleura.com","date":"2025-03-21 15:29:53.000000000","tz":0},"subject":"Use distro-specific SSH service name","message":"Use distro-specific SSH service name\n\nSome Ubuntu 24.04 images do not contain sshd service anymore,\nthus fail on the handler.\n\nLet\u0027s use distro-specific service name, like we do in other places.\n\nChange-Id: Ie5b0f215cceb8e8002b75543f4050042732525a7\n(cherry picked from commit 84b01a968460192c3fb18fa65630984b6ddf1123)\n","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/openstack-ansible-lxc_hosts/commit/3739ef843f99c40249ea859615fab1a33ae4bd70"}],"resolve_conflicts_web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/openstack-ansible-lxc_hosts/commit/3739ef843f99c40249ea859615fab1a33ae4bd70"}]},"branch":"refs/heads/stable/2024.2"},"8ea884c75e2c1803152337cb75cae92d009488d4":{"kind":"REWORK","_number":2,"created":"2025-03-21 15:30:45.000000000","uploader":{"_account_id":28619,"name":"Dmitriy Rabotyagov","email":"noonedeadpunk@gmail.com","username":"noonedeadpunk"},"ref":"refs/changes/29/945229/2","fetch":{"anonymous http":{"url":"https://review.opendev.org/openstack/openstack-ansible-lxc_hosts","ref":"refs/changes/29/945229/2","commands":{"Checkout":"git fetch https://review.opendev.org/openstack/openstack-ansible-lxc_hosts refs/changes/29/945229/2 \u0026\u0026 git checkout FETCH_HEAD","Cherry Pick":"git fetch https://review.opendev.org/openstack/openstack-ansible-lxc_hosts refs/changes/29/945229/2 \u0026\u0026 git cherry-pick FETCH_HEAD","Format Patch":"git fetch https://review.opendev.org/openstack/openstack-ansible-lxc_hosts refs/changes/29/945229/2 \u0026\u0026 git format-patch -1 --stdout FETCH_HEAD","Pull":"git pull https://review.opendev.org/openstack/openstack-ansible-lxc_hosts refs/changes/29/945229/2"}}},"commit":{"parents":[{"commit":"41470784fd9dc2438799f45417a9fbcef0d069d3","subject":"Update TOX_CONSTRAINTS_FILE for stable/2024.2","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/openstack-ansible-lxc_hosts/commit/41470784fd9dc2438799f45417a9fbcef0d069d3"}]}],"author":{"name":"Dmitriy Rabotyagov","email":"dmitriy.rabotyagov@cleura.com","date":"2025-02-26 14:33:52.000000000","tz":60},"committer":{"name":"Dmitriy Rabotyagov","email":"dmitriy.rabotyagov@cleura.com","date":"2025-03-21 15:30:32.000000000","tz":0},"subject":"Use distro-specific SSH service name","message":"Use distro-specific SSH service name\n\nSome Ubuntu 24.04 images do not contain sshd service anymore,\nthus fail on the handler.\n\nLet\u0027s use distro-specific service name, like we do in other places.\n\nChange-Id: Ie5b0f215cceb8e8002b75543f4050042732525a7\n(cherry picked from commit 84b01a968460192c3fb18fa65630984b6ddf1123)\n","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/openstack-ansible-lxc_hosts/commit/8ea884c75e2c1803152337cb75cae92d009488d4"}],"resolve_conflicts_web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/openstack-ansible-lxc_hosts/commit/8ea884c75e2c1803152337cb75cae92d009488d4"}]},"branch":"refs/heads/stable/2024.2"},"504024d047e066d767e9b71c10931d0c8c8d42ba":{"kind":"REWORK","_number":3,"created":"2025-03-21 15:31:32.000000000","uploader":{"_account_id":28619,"name":"Dmitriy Rabotyagov","email":"noonedeadpunk@gmail.com","username":"noonedeadpunk"},"ref":"refs/changes/29/945229/3","fetch":{"anonymous http":{"url":"https://review.opendev.org/openstack/openstack-ansible-lxc_hosts","ref":"refs/changes/29/945229/3","commands":{"Checkout":"git fetch https://review.opendev.org/openstack/openstack-ansible-lxc_hosts refs/changes/29/945229/3 \u0026\u0026 git checkout FETCH_HEAD","Cherry Pick":"git fetch https://review.opendev.org/openstack/openstack-ansible-lxc_hosts refs/changes/29/945229/3 \u0026\u0026 git cherry-pick FETCH_HEAD","Format Patch":"git fetch https://review.opendev.org/openstack/openstack-ansible-lxc_hosts refs/changes/29/945229/3 \u0026\u0026 git format-patch -1 --stdout FETCH_HEAD","Pull":"git pull https://review.opendev.org/openstack/openstack-ansible-lxc_hosts refs/changes/29/945229/3"}}},"commit":{"parents":[{"commit":"41470784fd9dc2438799f45417a9fbcef0d069d3","subject":"Update TOX_CONSTRAINTS_FILE for stable/2024.2","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/openstack-ansible-lxc_hosts/commit/41470784fd9dc2438799f45417a9fbcef0d069d3"}]}],"author":{"name":"Dmitriy Rabotyagov","email":"dmitriy.rabotyagov@cleura.com","date":"2025-02-26 14:33:52.000000000","tz":60},"committer":{"name":"Dmitriy Rabotyagov","email":"dmitriy.rabotyagov@cleura.com","date":"2025-03-21 15:31:31.000000000","tz":0},"subject":"Use distro-specific SSH service name","message":"Use distro-specific SSH service name\n\nSome Ubuntu 24.04 images do not contain sshd service anymore,\nthus fail on the handler.\n\nLet\u0027s use distro-specific service name, like we do in other places.\n\nChange-Id: Ie5b0f215cceb8e8002b75543f4050042732525a7\n(cherry picked from commit 84b01a968460192c3fb18fa65630984b6ddf1123)\n","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/openstack-ansible-lxc_hosts/commit/504024d047e066d767e9b71c10931d0c8c8d42ba"}],"resolve_conflicts_web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/openstack-ansible-lxc_hosts/commit/504024d047e066d767e9b71c10931d0c8c8d42ba"}]},"branch":"refs/heads/stable/2024.2"}},"requirements":[],"submit_records":[{"rule_name":"gerrit~DefaultSubmitRule","status":"CLOSED","labels":[{"label":"Verified","status":"MAY","applied_by":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]}},{"label":"Code-Review","status":"MAY","applied_by":{"_account_id":28619,"name":"Dmitriy Rabotyagov","email":"noonedeadpunk@gmail.com","username":"noonedeadpunk"}},{"label":"Workflow","status":"MAY","applied_by":{"_account_id":25023,"name":"Jonathan Rosser","email":"jonathan.rosser@rd.bbc.co.uk","username":"jrosser"}},{"label":"Backport-Candidate","status":"MAY"}]}],"submit_requirements":[{"name":"Verified","description":"Verified in gate by CI","status":"SATISFIED","is_legacy":false,"submittability_expression_result":{"expression":"label:Verified\u003dMAX AND -label:Verified\u003dMIN","fulfilled":true,"status":"PASS","passing_atoms":["label:Verified\u003dMAX"],"failing_atoms":["label:Verified\u003dMIN"],"atom_explanations":{}}},{"name":"Backport-Candidate","description":"Backport candidate status","status":"NOT_APPLICABLE","is_legacy":false,"applicability_expression_result":{"fulfilled":false,"status":"FAIL"},"submittability_expression_result":{"expression":"is:true","fulfilled":true,"status":"NOT_EVALUATED","passing_atoms":[],"failing_atoms":[],"atom_explanations":{}}},{"name":"Code-Review","description":"Code reviewed by core reviewer","status":"SATISFIED","is_legacy":false,"submittability_expression_result":{"expression":"label:Code-Review\u003dMAX AND -label:Code-Review\u003dMIN","fulfilled":true,"status":"PASS","passing_atoms":["label:Code-Review\u003dMAX"],"failing_atoms":["label:Code-Review\u003dMIN"],"atom_explanations":{}}},{"name":"Workflow","description":"Approved for gate by core reviewer","status":"SATISFIED","is_legacy":false,"submittability_expression_result":{"expression":"label:Workflow\u003dMAX AND -label:Workflow\u003dMIN","fulfilled":true,"status":"PASS","passing_atoms":["label:Workflow\u003dMAX"],"failing_atoms":["label:Workflow\u003dMIN"],"atom_explanations":{}}}]}
