)]}'
{"id":"openstack%2Fopenstack-ansible-lxc_hosts~949848","triplet_id":"openstack%2Fopenstack-ansible-lxc_hosts~stable%2F2024.2~I3de2c45fc1b24424ccbb8035cc7e7603dc5d0976","project":"openstack/openstack-ansible-lxc_hosts","branch":"stable/2024.2","attention_set":{},"removed_from_attention_set":{"28619":{"account":{"_account_id":28619,"name":"Dmitriy Rabotyagov","email":"noonedeadpunk@gmail.com","username":"noonedeadpunk"},"last_update":"2025-05-16 18:39:03.000000000","reason":"Change was submitted"}},"hashtags":[],"change_id":"I3de2c45fc1b24424ccbb8035cc7e7603dc5d0976","subject":"Patch the usr.bin.lxc-copy apparmor profile","status":"MERGED","created":"2025-05-15 06:43:36.000000000","updated":"2025-05-16 18:39:47.000000000","submitted":"2025-05-16 18:39:03.000000000","submitter":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"total_comment_count":2,"unresolved_comment_count":0,"has_review_started":true,"submission_id":"949848","cherry_pick_of_change":949740,"cherry_pick_of_patch_set":3,"meta_rev_id":"d65239d7551f0986599656ff8195702e0ca00b46","_number":949848,"virtual_id_number":949848,"owner":{"_account_id":28619,"name":"Dmitriy Rabotyagov","email":"noonedeadpunk@gmail.com","username":"noonedeadpunk"},"actions":{},"labels":{"Verified":{"approved":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"all":[{"value":0,"_account_id":28619,"name":"Dmitriy Rabotyagov","email":"noonedeadpunk@gmail.com","username":"noonedeadpunk"},{"tag":"autogenerated:zuul:gate","value":2,"date":"2025-05-16 18:39:03.000000000","permitted_voting_range":{"min":2,"max":2},"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},{"value":0,"_account_id":25023,"name":"Jonathan Rosser","email":"jonathan.rosser@rd.bbc.co.uk","username":"jrosser"}],"values":{"-2":"Fails","-1":"Doesn\u0027t seem to work"," 0":"No score","+1":"Works for me","+2":"Verified"},"description":"","default_value":0,"optional":true},"Code-Review":{"approved":{"_account_id":28619,"name":"Dmitriy Rabotyagov","email":"noonedeadpunk@gmail.com","username":"noonedeadpunk"},"all":[{"value":2,"date":"2025-05-15 10:41:32.000000000","permitted_voting_range":{"min":2,"max":2},"_account_id":28619,"name":"Dmitriy Rabotyagov","email":"noonedeadpunk@gmail.com","username":"noonedeadpunk"},{"value":0,"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},{"value":2,"date":"2025-05-16 16:49:14.000000000","permitted_voting_range":{"min":2,"max":2},"_account_id":25023,"name":"Jonathan Rosser","email":"jonathan.rosser@rd.bbc.co.uk","username":"jrosser"}],"values":{"-2":"Do not merge","-1":"This patch needs further work before it can be merged"," 0":"No score","+1":"Looks good to me, but someone else must approve","+2":"Looks good to me (core reviewer)"},"description":"","default_value":0,"optional":true},"Workflow":{"approved":{"_account_id":25023,"name":"Jonathan Rosser","email":"jonathan.rosser@rd.bbc.co.uk","username":"jrosser"},"all":[{"value":0,"_account_id":28619,"name":"Dmitriy Rabotyagov","email":"noonedeadpunk@gmail.com","username":"noonedeadpunk"},{"value":0,"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},{"value":1,"date":"2025-05-16 16:49:14.000000000","permitted_voting_range":{"min":1,"max":1},"_account_id":25023,"name":"Jonathan Rosser","email":"jonathan.rosser@rd.bbc.co.uk","username":"jrosser"}],"values":{"-1":"Work in progress"," 0":"Ready for reviews","+1":"Approved"},"description":"","default_value":0,"optional":true},"Backport-Candidate":{"all":[{"value":0,"_account_id":28619,"name":"Dmitriy Rabotyagov","email":"noonedeadpunk@gmail.com","username":"noonedeadpunk"},{"value":0,"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},{"value":0,"_account_id":25023,"name":"Jonathan Rosser","email":"jonathan.rosser@rd.bbc.co.uk","username":"jrosser"}],"values":{"-1":"Not A Backport Candidate"," 0":"Backport Review Needed","+1":"Proposed Backport"},"description":"","default_value":0,"optional":true}},"removable_reviewers":[],"reviewers":{"REVIEWER":[{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},{"_account_id":25023,"name":"Jonathan Rosser","email":"jonathan.rosser@rd.bbc.co.uk","username":"jrosser"},{"_account_id":28619,"name":"Dmitriy Rabotyagov","email":"noonedeadpunk@gmail.com","username":"noonedeadpunk"}]},"pending_reviewers":{},"reviewer_updates":[{"updated":"2025-05-15 08:23:24.000000000","updated_by":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"reviewer":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"state":"REVIEWER"},{"updated":"2025-05-16 07:35:05.000000000","updated_by":{"_account_id":25023,"name":"Jonathan Rosser","email":"jonathan.rosser@rd.bbc.co.uk","username":"jrosser"},"reviewer":{"_account_id":25023,"name":"Jonathan Rosser","email":"jonathan.rosser@rd.bbc.co.uk","username":"jrosser"},"state":"CC"},{"updated":"2025-05-16 16:49:14.000000000","updated_by":{"_account_id":25023,"name":"Jonathan Rosser","email":"jonathan.rosser@rd.bbc.co.uk","username":"jrosser"},"reviewer":{"_account_id":25023,"name":"Jonathan Rosser","email":"jonathan.rosser@rd.bbc.co.uk","username":"jrosser"},"state":"REVIEWER"}],"messages":[{"id":"5d64a65bc09c340509f9bcc339b4145d514ba25e","tag":"autogenerated:gerrit:newPatchSet","author":{"_account_id":28619,"name":"Dmitriy Rabotyagov","email":"noonedeadpunk@gmail.com","username":"noonedeadpunk"},"date":"2025-05-15 06:43:36.000000000","message":"Patch Set 1: Cherry Picked from branch master.","accounts_in_message":[],"_revision_number":1},{"id":"946b994c67b85aec457ca351ed427691c1500188","tag":"autogenerated:zuul:check","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2025-05-15 08:23:24.000000000","message":"Patch Set 1: Verified-1\n\nBuild failed (check pipeline).  For information on how to proceed, see\nhttps://docs.opendev.org/opendev/infra-manual/latest/developers.html#automated-testing\nand https://docs.openstack.org/project-team-guide/testing.html#how-to-handle-test-failures\n\nhttps://zuul.opendev.org/t/openstack/buildset/73ac284e8a7d42579992d31839820d79\n\n- openstack-ansible-linters-ubuntu-noble https://zuul.opendev.org/t/openstack/build/6d853dbd2f2a4b45a52f30ada8aaeb56 : SUCCESS in 8m 43s\n- openstack-tox-docs https://zuul.opendev.org/t/openstack/build/69a05d5ae7504989a87b331762704e92 : SUCCESS in 5m 11s\n- build-openstack-releasenotes https://zuul.opendev.org/t/openstack/build/a9ff880d51ae4015ace1dfd24fe4607b : SUCCESS in 2m 30s\n- openstack-ansible-deploy-infra_lxc-centos-9-stream https://zuul.opendev.org/t/openstack/build/d914fc4d96cc4c36bdc4d4c67d585433 : SUCCESS in 1h 15m 08s (non-voting)\n- openstack-ansible-deploy-infra_lxc-rockylinux-9 https://zuul.opendev.org/t/openstack/build/5f20b50768714a668a30ee3f4f1ac2e1 : SUCCESS in 1h 26m 51s\n- openstack-ansible-deploy-infra_lxc_tls-rockylinux-9 https://zuul.opendev.org/t/openstack/build/de9c7a2f6cab45b1b00ef82119900700 : SUCCESS in 43m 57s\n- openstack-ansible-deploy-infra_lxc-debian-bookworm https://zuul.opendev.org/t/openstack/build/9e602a1d4ba949da9ad61f35e003c312 : SUCCESS in 49m 12s\n- openstack-ansible-deploy-infra_lxc-ubuntu-jammy https://zuul.opendev.org/t/openstack/build/044691652ca0473484442fce04afb15f : SUCCESS in 1h 10m 40s\n- openstack-ansible-deploy-infra_lxc_stepca-ubuntu-jammy https://zuul.opendev.org/t/openstack/build/d87a685813024ce7ac11d4e86e920eca : FAILURE in 31m 48s\n- openstack-ansible-deploy-infra_lxc_tls-ubuntu-jammy https://zuul.opendev.org/t/openstack/build/9daaf8f23b08481aa90db596e17cc64d : SUCCESS in 1h 10m 34s\n- openstack-ansible-deploy-infra_lxc-ubuntu-noble https://zuul.opendev.org/t/openstack/build/6228b7fac0c849e39b66a4e7bc561ae5 : SUCCESS in 50m 16s\n- openstack-ansible-deploy-infra_lxc_stepca-ubuntu-noble https://zuul.opendev.org/t/openstack/build/8d3aa189deae48788f38932c339c041f : SUCCESS in 43m 14s\n- openstack-ansible-deploy-infra_lxc_tls-ubuntu-noble https://zuul.opendev.org/t/openstack/build/3867dae3a6824f23870642d3bcf1d6da : SUCCESS in 43m 16s\n- openstack-ansible-upgrade-infra_lxc-rockylinux-9 https://zuul.opendev.org/t/openstack/build/0f86edc16065404f81652ca5e3b53598 : SUCCESS in 1h 08m 06s\n- openstack-ansible-upgrade-infra_lxc-ubuntu-jammy https://zuul.opendev.org/t/openstack/build/8e09142cfef74b74b37d2f359a7f3eb6 : SUCCESS in 1h 37m 34s\n- openstack-ansible-deploy-hosts_lxc_zfs-ubuntu-jammy https://zuul.opendev.org/t/openstack/build/5406e8b22da1429db154f1bce39228c5 : SUCCESS in 21m 15s\n- openstack-ansible-deploy-hosts_lxc_btrfs-ubuntu-jammy https://zuul.opendev.org/t/openstack/build/92f257e9b75f47fca5c891dec27ccc20 : SUCCESS in 21m 09s\n- openstack-ansible-deploy-hosts_lxc_overlayfs-ubuntu-jammy https://zuul.opendev.org/t/openstack/build/e050786ef21b4947900248362bc1ffac : SUCCESS in 21m 29s\n- openstack-ansible-deploy-hosts_lxc_zfs-ubuntu-noble https://zuul.opendev.org/t/openstack/build/4c40a57b9b634542a06a7139a1aac6e4 : SUCCESS in 21m 58s\n- openstack-ansible-deploy-hosts_lxc_btrfs-ubuntu-noble https://zuul.opendev.org/t/openstack/build/cdab1c18f3c14008ba8560f36733deb8 : SUCCESS in 23m 03s\n- openstack-ansible-deploy-hosts_lxc_overlayfs-ubuntu-noble https://zuul.opendev.org/t/openstack/build/b3e4f2613e5040a6ae7bc6ac4c5d7c03 : SUCCESS in 22m 30s\n- openstack-ansible-deploy-hosts_lxc_btrfs-debian-bookworm https://zuul.opendev.org/t/openstack/build/41a79d83bb7a45c39743b2f03202b846 : SUCCESS in 20m 39s\n- openstack-ansible-deploy-hosts_lxc_overlayfs-debian-bookworm https://zuul.opendev.org/t/openstack/build/467827f21ca543b38eb69b539ffce43a : SUCCESS in 21m 05s\n- openstack-ansible-deploy-hosts_lxc_overlayfs-rockylinux-9 https://zuul.opendev.org/t/openstack/build/5c25744619674818bb276fcb4198a7ba : SUCCESS in 23m 08s","accounts_in_message":[],"_revision_number":1},{"id":"0738659963fc83b72a0065ddd4bf369afae9181b","author":{"_account_id":28619,"name":"Dmitriy Rabotyagov","email":"noonedeadpunk@gmail.com","username":"noonedeadpunk"},"date":"2025-05-15 10:10:41.000000000","message":"Patch Set 1:\n\n(1 comment)","accounts_in_message":[],"_revision_number":1},{"id":"2caa53dfc558bdd1135fd2f6b4a9340803994d63","author":{"_account_id":28619,"name":"Dmitriy Rabotyagov","email":"noonedeadpunk@gmail.com","username":"noonedeadpunk"},"date":"2025-05-15 10:41:32.000000000","message":"Patch Set 1: Code-Review+2","accounts_in_message":[],"_revision_number":1},{"id":"bd4513a6ca1786129275746650eab3b358da6af5","tag":"autogenerated:zuul:check","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2025-05-15 11:45:58.000000000","message":"Patch Set 1:\n\nBuild failed (check pipeline).  For information on how to proceed, see\nhttps://docs.opendev.org/opendev/infra-manual/latest/developers.html#automated-testing\nand https://docs.openstack.org/project-team-guide/testing.html#how-to-handle-test-failures\n\nhttps://zuul.opendev.org/t/openstack/buildset/819f8a8df05f4818b32809c2cb391e48\n\n- openstack-ansible-linters-ubuntu-noble https://zuul.opendev.org/t/openstack/build/bedee1d86d764ec6a99cb17b7770573a : SUCCESS in 4m 44s\n- openstack-tox-docs https://zuul.opendev.org/t/openstack/build/1c01c3e4b64c4951bbd0b5dd763205bc : SUCCESS in 2m 35s\n- build-openstack-releasenotes https://zuul.opendev.org/t/openstack/build/2d04e20815fa47619518c168201eeb81 : SUCCESS in 1m 26s\n- openstack-ansible-deploy-infra_lxc-centos-9-stream https://zuul.opendev.org/t/openstack/build/d43ecb9f603346ce9f80ca7d2aa36425 : SUCCESS in 1h 24m 21s (non-voting)\n- openstack-ansible-deploy-infra_lxc-rockylinux-9 https://zuul.opendev.org/t/openstack/build/88e36243b0f44b929f7083e079ced135 : SUCCESS in 1h 16m 40s\n- openstack-ansible-deploy-infra_lxc_tls-rockylinux-9 https://zuul.opendev.org/t/openstack/build/48b5f8ec46b94dbe9a7dfd6076b63985 : SUCCESS in 1h 28m 08s\n- openstack-ansible-deploy-infra_lxc-debian-bookworm https://zuul.opendev.org/t/openstack/build/149bbb423b1f4a56bf1b86e2f5bad435 : SUCCESS in 1h 10m 29s\n- openstack-ansible-deploy-infra_lxc-ubuntu-jammy https://zuul.opendev.org/t/openstack/build/08e9e8d55de14ee3876198e864b58cc9 : SUCCESS in 1h 16m 19s\n- openstack-ansible-deploy-infra_lxc_stepca-ubuntu-jammy https://zuul.opendev.org/t/openstack/build/fc30bfd1158b49afa645a09e3fbde33d : SUCCESS in 1h 09m 39s\n- openstack-ansible-deploy-infra_lxc_tls-ubuntu-jammy https://zuul.opendev.org/t/openstack/build/97c2aaaac83249649ce997bc9f8ba30a : SUCCESS in 1h 17m 37s\n- openstack-ansible-deploy-infra_lxc-ubuntu-noble https://zuul.opendev.org/t/openstack/build/a2f5316b59504b89b8ddbdb1a1950b46 : SUCCESS in 1h 13m 17s\n- openstack-ansible-deploy-infra_lxc_stepca-ubuntu-noble https://zuul.opendev.org/t/openstack/build/373843d454dd4db6948f6aac5f2c2e86 : SUCCESS in 1h 21m 40s\n- openstack-ansible-deploy-infra_lxc_tls-ubuntu-noble https://zuul.opendev.org/t/openstack/build/45036042db5445d2b69274a55ba07222 : SUCCESS in 1h 08m 22s\n- openstack-ansible-upgrade-infra_lxc-rockylinux-9 https://zuul.opendev.org/t/openstack/build/37b0ee7aeebf4492954bf1c77d15356b : SUCCESS in 1h 23m 18s\n- openstack-ansible-upgrade-infra_lxc-ubuntu-jammy https://zuul.opendev.org/t/openstack/build/20e85b358f054e958f042aec23d3f77d : FAILURE in 1h 09m 49s\n- openstack-ansible-deploy-hosts_lxc_zfs-ubuntu-jammy https://zuul.opendev.org/t/openstack/build/8804b9dacc604bd4bcf8186c8758c497 : SUCCESS in 18m 14s\n- openstack-ansible-deploy-hosts_lxc_btrfs-ubuntu-jammy https://zuul.opendev.org/t/openstack/build/d40b12d39cbd4047bc32595c48443feb : SUCCESS in 20m 03s\n- openstack-ansible-deploy-hosts_lxc_overlayfs-ubuntu-jammy https://zuul.opendev.org/t/openstack/build/87bafbe32b354fc4be85ce816eed1673 : SUCCESS in 17m 12s\n- openstack-ansible-deploy-hosts_lxc_zfs-ubuntu-noble https://zuul.opendev.org/t/openstack/build/2db3a7a180c147f6833badc06958f142 : SUCCESS in 18m 08s\n- openstack-ansible-deploy-hosts_lxc_btrfs-ubuntu-noble https://zuul.opendev.org/t/openstack/build/e9fa58a36aaf4c8fa5785be8ae277694 : SUCCESS in 19m 00s\n- openstack-ansible-deploy-hosts_lxc_overlayfs-ubuntu-noble https://zuul.opendev.org/t/openstack/build/9609025f6c384e9b9a0efadcb9e40f20 : SUCCESS in 20m 23s\n- openstack-ansible-deploy-hosts_lxc_btrfs-debian-bookworm https://zuul.opendev.org/t/openstack/build/d679ad641c96424e9476a5670c30ffef : SUCCESS in 19m 24s\n- openstack-ansible-deploy-hosts_lxc_overlayfs-debian-bookworm https://zuul.opendev.org/t/openstack/build/cf56793f61b84d9eb517460d3b1843e4 : SUCCESS in 19m 30s\n- openstack-ansible-deploy-hosts_lxc_overlayfs-rockylinux-9 https://zuul.opendev.org/t/openstack/build/6a6375c4f1234135bb9588fe4d2e4cbd : SUCCESS in 14m 23s","accounts_in_message":[],"_revision_number":1},{"id":"1f2447e790a8995ef80357ed15bfccf1491b33cc","author":{"_account_id":25023,"name":"Jonathan Rosser","email":"jonathan.rosser@rd.bbc.co.uk","username":"jrosser"},"date":"2025-05-16 07:35:05.000000000","message":"Patch Set 1:\n\n(1 comment)","accounts_in_message":[],"_revision_number":1},{"id":"9caba8efa9a4befffbb1ec5b18632bfeef5a0143","tag":"autogenerated:zuul:check","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2025-05-16 09:34:31.000000000","message":"Patch Set 1: Verified+1\n\nBuild succeeded (check pipeline).\nhttps://zuul.opendev.org/t/openstack/buildset/fd486f7e8b0a42439814f7cca8ed4210\n\n- openstack-ansible-linters-ubuntu-noble https://zuul.opendev.org/t/openstack/build/7dcc878de6fc474288fb47e487ed1e77 : SUCCESS in 6m 03s\n- openstack-tox-docs https://zuul.opendev.org/t/openstack/build/8bc5e963262d432f8ff10ead63d7f89c : SUCCESS in 5m 59s\n- build-openstack-releasenotes https://zuul.opendev.org/t/openstack/build/c860f71f01c641ee83cfc92438f9cd7c : SUCCESS in 2m 39s\n- openstack-ansible-deploy-infra_lxc-centos-9-stream https://zuul.opendev.org/t/openstack/build/4e98550e3f0d4f47b5a858aeab85c98c : SUCCESS in 42m 33s (non-voting)\n- openstack-ansible-deploy-infra_lxc-rockylinux-9 https://zuul.opendev.org/t/openstack/build/bb45fa89e9064d3ea9216c50d816b08f : SUCCESS in 1h 17m 38s\n- openstack-ansible-deploy-infra_lxc_tls-rockylinux-9 https://zuul.opendev.org/t/openstack/build/e9a20d6ae7a642529017f06e8df645fd : SUCCESS in 1h 26m 34s\n- openstack-ansible-deploy-infra_lxc-debian-bookworm https://zuul.opendev.org/t/openstack/build/a4068a76e7eb4595ad16358c2ab03f54 : SUCCESS in 40m 15s\n- openstack-ansible-deploy-infra_lxc-ubuntu-jammy https://zuul.opendev.org/t/openstack/build/c6b902f5e1d343e1bc3ecd54864b394e : SUCCESS in 1h 07m 58s\n- openstack-ansible-deploy-infra_lxc_stepca-ubuntu-jammy https://zuul.opendev.org/t/openstack/build/f2072153191245869b09ba8b4e1ad97e : SUCCESS in 1h 13m 20s\n- openstack-ansible-deploy-infra_lxc_tls-ubuntu-jammy https://zuul.opendev.org/t/openstack/build/3156d96176e8437d8eb61c5c865a5417 : SUCCESS in 1h 14m 35s\n- openstack-ansible-deploy-infra_lxc-ubuntu-noble https://zuul.opendev.org/t/openstack/build/0855dcc5c0604b6d9f6d150d2d801b43 : SUCCESS in 1h 17m 20s\n- openstack-ansible-deploy-infra_lxc_stepca-ubuntu-noble https://zuul.opendev.org/t/openstack/build/946b483f11fa4ec5bd4f89a5b416792f : SUCCESS in 1h 08m 46s\n- openstack-ansible-deploy-infra_lxc_tls-ubuntu-noble https://zuul.opendev.org/t/openstack/build/ee581b09460f43d98725d2b0dc27d0aa : SUCCESS in 43m 05s\n- openstack-ansible-upgrade-infra_lxc-rockylinux-9 https://zuul.opendev.org/t/openstack/build/2ecf5492a7164ba7ab2728c64b4d9c74 : SUCCESS in 1h 51m 31s\n- openstack-ansible-upgrade-infra_lxc-ubuntu-jammy https://zuul.opendev.org/t/openstack/build/fa09120c5194408a9d80340286dc6e1e : SUCCESS in 1h 38m 17s\n- openstack-ansible-deploy-hosts_lxc_zfs-ubuntu-jammy https://zuul.opendev.org/t/openstack/build/397fb25c16b5440989cd5a141a1961a6 : SUCCESS in 20m 05s\n- openstack-ansible-deploy-hosts_lxc_btrfs-ubuntu-jammy https://zuul.opendev.org/t/openstack/build/d0aa9e03fd30469280f12fd9c3f5fe65 : SUCCESS in 19m 03s\n- openstack-ansible-deploy-hosts_lxc_overlayfs-ubuntu-jammy https://zuul.opendev.org/t/openstack/build/e656b6a8b0ab45ed8928618f1e256503 : SUCCESS in 21m 39s\n- openstack-ansible-deploy-hosts_lxc_zfs-ubuntu-noble https://zuul.opendev.org/t/openstack/build/8811908654e04b1d9fe87fbb5340506e : SUCCESS in 20m 21s\n- openstack-ansible-deploy-hosts_lxc_btrfs-ubuntu-noble https://zuul.opendev.org/t/openstack/build/2d806326fdd342ab943691b32a53771b : SUCCESS in 23m 48s\n- openstack-ansible-deploy-hosts_lxc_overlayfs-ubuntu-noble https://zuul.opendev.org/t/openstack/build/14ad364fa6c7416f92bcd4e5ff4c6531 : SUCCESS in 20m 40s\n- openstack-ansible-deploy-hosts_lxc_btrfs-debian-bookworm https://zuul.opendev.org/t/openstack/build/e088b8a2c030413d9a26aa9ada6a219f : SUCCESS in 23m 25s\n- openstack-ansible-deploy-hosts_lxc_overlayfs-debian-bookworm https://zuul.opendev.org/t/openstack/build/62cfb7fd4d9a42ed8921ab9233d03465 : SUCCESS in 22m 58s\n- openstack-ansible-deploy-hosts_lxc_overlayfs-rockylinux-9 https://zuul.opendev.org/t/openstack/build/ee23f18b47064dabbf9f8c1871a658b9 : SUCCESS in 20m 20s","accounts_in_message":[],"_revision_number":1},{"id":"970fb87b13f48af1e1b8cd98c9069fc5dfd75092","author":{"_account_id":25023,"name":"Jonathan Rosser","email":"jonathan.rosser@rd.bbc.co.uk","username":"jrosser"},"date":"2025-05-16 16:49:14.000000000","message":"Patch Set 1: Code-Review+2 Workflow+1","accounts_in_message":[],"_revision_number":1},{"id":"b74ce4d775441809dd6977daa0932536f3a652f9","tag":"autogenerated:zuul:gate","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2025-05-16 16:51:03.000000000","message":"Patch Set 1: -Verified\n\nStarting gate jobs.","accounts_in_message":[],"_revision_number":1},{"id":"42143ffb13113c782a752a11ea7e4dd431f8332c","tag":"autogenerated:zuul:gate","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2025-05-16 18:39:03.000000000","message":"Patch Set 1: Verified+2\n\nBuild succeeded (gate pipeline).\nhttps://zuul.opendev.org/t/openstack/buildset/9058b741f8dd45d6ae1486923e8e26ad\n\n- openstack-ansible-linters-ubuntu-noble https://zuul.opendev.org/t/openstack/build/1c119652a63e40829b1c11c67a8472a1 : SUCCESS in 6m 21s\n- openstack-tox-docs https://zuul.opendev.org/t/openstack/build/0a0b9d653374495ca51536c6f3554548 : SUCCESS in 4m 05s\n- build-openstack-releasenotes https://zuul.opendev.org/t/openstack/build/a63a6d8c2d9742ada04fd82d598d6473 : SUCCESS in 2m 42s\n- openstack-ansible-deploy-infra_lxc-rockylinux-9 https://zuul.opendev.org/t/openstack/build/688f371550e74e75a16a786d7697f738 : SUCCESS in 1h 21m 52s\n- openstack-ansible-deploy-infra_lxc_tls-rockylinux-9 https://zuul.opendev.org/t/openstack/build/005c64eab8344d1f883fc6d37b5d796f : SUCCESS in 1h 17m 48s\n- openstack-ansible-deploy-infra_lxc-debian-bookworm https://zuul.opendev.org/t/openstack/build/3344960b26db4e4e894b616fe4761c47 : SUCCESS in 1h 11m 27s\n- openstack-ansible-deploy-infra_lxc-ubuntu-jammy https://zuul.opendev.org/t/openstack/build/a9a70e009b044968ad7557f0daa5550b : SUCCESS in 1h 07m 24s\n- openstack-ansible-deploy-infra_lxc_stepca-ubuntu-jammy https://zuul.opendev.org/t/openstack/build/acf418103d484c289d5a4ecb7f9132ed : SUCCESS in 39m 08s\n- openstack-ansible-deploy-infra_lxc_tls-ubuntu-jammy https://zuul.opendev.org/t/openstack/build/d1ec6ac095fc4c9a85a9ad713d7e3704 : SUCCESS in 47m 08s\n- openstack-ansible-deploy-infra_lxc-ubuntu-noble https://zuul.opendev.org/t/openstack/build/8b70367c1325410a940a9d3d3755eb34 : SUCCESS in 1h 12m 35s\n- openstack-ansible-deploy-infra_lxc_stepca-ubuntu-noble https://zuul.opendev.org/t/openstack/build/375b990be0394fecbd7b045ad85e020b : SUCCESS in 1h 08m 18s\n- openstack-ansible-deploy-infra_lxc_tls-ubuntu-noble https://zuul.opendev.org/t/openstack/build/255e968b8cbb434bb6352a719cc57ce6 : SUCCESS in 1h 19m 04s\n- openstack-ansible-upgrade-infra_lxc-rockylinux-9 https://zuul.opendev.org/t/openstack/build/ff02492d896e48a7bd654c5f7f6d6168 : SUCCESS in 1h 39m 53s\n- openstack-ansible-upgrade-infra_lxc-ubuntu-jammy https://zuul.opendev.org/t/openstack/build/e0253ecf3033452db34143033e5262fd : SUCCESS in 1h 43m 33s\n- openstack-ansible-deploy-hosts_lxc_zfs-ubuntu-jammy https://zuul.opendev.org/t/openstack/build/3ac9f771df5847439392fd293bd75067 : SUCCESS in 12m 41s\n- openstack-ansible-deploy-hosts_lxc_btrfs-ubuntu-jammy https://zuul.opendev.org/t/openstack/build/baf6b2a76fa04f7ea60a4bf722df5979 : SUCCESS in 14m 08s\n- openstack-ansible-deploy-hosts_lxc_overlayfs-ubuntu-jammy https://zuul.opendev.org/t/openstack/build/621980ac7a1343a3b7371fba5956f0f2 : SUCCESS in 20m 51s\n- openstack-ansible-deploy-hosts_lxc_zfs-ubuntu-noble https://zuul.opendev.org/t/openstack/build/9bdef9211e8545eeb069dcff98e12785 : SUCCESS in 22m 44s\n- openstack-ansible-deploy-hosts_lxc_btrfs-ubuntu-noble https://zuul.opendev.org/t/openstack/build/936da4127ce141358ca0b2febccb35be : SUCCESS in 26m 20s\n- openstack-ansible-deploy-hosts_lxc_overlayfs-ubuntu-noble https://zuul.opendev.org/t/openstack/build/fc64af850d6848758aeb222182d38506 : SUCCESS in 20m 18s\n- openstack-ansible-deploy-hosts_lxc_btrfs-debian-bookworm https://zuul.opendev.org/t/openstack/build/97bbdad31b0b4805b1e4ed6cd478d86e : SUCCESS in 21m 12s\n- openstack-ansible-deploy-hosts_lxc_overlayfs-debian-bookworm https://zuul.opendev.org/t/openstack/build/21e8f609d40a4bb69e8e548d0de1c390 : SUCCESS in 25m 07s\n- openstack-ansible-deploy-hosts_lxc_overlayfs-rockylinux-9 https://zuul.opendev.org/t/openstack/build/0fea2d1cdc3f4333a5c0ba1f726a6b25 : SUCCESS in 26m 13s","accounts_in_message":[],"_revision_number":1},{"id":"ec42d76e7aee29adcaf81821703abb6194aadb52","tag":"autogenerated:gerrit:merged","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2025-05-16 18:39:03.000000000","message":"Change has been successfully merged","accounts_in_message":[],"_revision_number":1},{"id":"d65239d7551f0986599656ff8195702e0ca00b46","tag":"autogenerated:zuul:promote","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2025-05-16 18:39:47.000000000","message":"Patch Set 1:\n\nBuild succeeded (promote pipeline).\nhttps://zuul.opendev.org/t/openstack/buildset/da92164eec3848a5bd26aeecc8ae2751\n\n- promote-openstack-tox-docs https://zuul.opendev.org/t/openstack/build/6629e4aa3bf64ad2a069fb1666383482 : SUCCESS in 37s","accounts_in_message":[],"_revision_number":1}],"current_revision_number":1,"current_revision":"a58cde80d982496a0dd0b4ad35f101ebb0c8db23","revisions":{"a58cde80d982496a0dd0b4ad35f101ebb0c8db23":{"kind":"REWORK","_number":1,"created":"2025-05-15 06:43:36.000000000","uploader":{"_account_id":28619,"name":"Dmitriy Rabotyagov","email":"noonedeadpunk@gmail.com","username":"noonedeadpunk"},"ref":"refs/changes/48/949848/1","fetch":{"anonymous http":{"url":"https://review.opendev.org/openstack/openstack-ansible-lxc_hosts","ref":"refs/changes/48/949848/1","commands":{"Checkout":"git fetch https://review.opendev.org/openstack/openstack-ansible-lxc_hosts refs/changes/48/949848/1 \u0026\u0026 git checkout FETCH_HEAD","Cherry Pick":"git fetch https://review.opendev.org/openstack/openstack-ansible-lxc_hosts refs/changes/48/949848/1 \u0026\u0026 git cherry-pick FETCH_HEAD","Format Patch":"git fetch https://review.opendev.org/openstack/openstack-ansible-lxc_hosts refs/changes/48/949848/1 \u0026\u0026 git format-patch -1 --stdout FETCH_HEAD","Pull":"git pull https://review.opendev.org/openstack/openstack-ansible-lxc_hosts refs/changes/48/949848/1"}}},"commit":{"parents":[{"commit":"504024d047e066d767e9b71c10931d0c8c8d42ba","subject":"Use distro-specific SSH service name","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/openstack-ansible-lxc_hosts/commit/504024d047e066d767e9b71c10931d0c8c8d42ba"}]}],"author":{"name":"Dmitriy Rabotyagov","email":"dmitriy.rabotyagov@cleura.com","date":"2025-05-14 12:57:58.000000000","tz":120},"committer":{"name":"Dmitriy Rabotyagov","email":"dmitriy.rabotyagov@cleura.com","date":"2025-05-15 06:43:36.000000000","tz":0},"subject":"Patch the usr.bin.lxc-copy apparmor profile","message":"Patch the usr.bin.lxc-copy apparmor profile\n\nDue to the bug in liblxc-common packaging [1], there is a conflict in\napparmor profiles which prevents any management of existing\nprofiles. In order to allow apparmor configuration we need to\nbushfix the usr.bin.lxc-copy profile until fixed one won\u0027t be released.\n\n[1] https://bugs.launchpad.net/ubuntu/+source/lxc/+bug/2110635\n\nChange-Id: I3de2c45fc1b24424ccbb8035cc7e7603dc5d0976\n(cherry picked from commit f059f97ce2a04f1618b53f68d78c9f37f3c699f1)\n","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/openstack-ansible-lxc_hosts/commit/a58cde80d982496a0dd0b4ad35f101ebb0c8db23"}],"resolve_conflicts_web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/openstack-ansible-lxc_hosts/commit/a58cde80d982496a0dd0b4ad35f101ebb0c8db23"}]},"branch":"refs/heads/stable/2024.2"}},"requirements":[],"submit_records":[{"rule_name":"gerrit~DefaultSubmitRule","status":"CLOSED","labels":[{"label":"Verified","status":"MAY","applied_by":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]}},{"label":"Code-Review","status":"MAY","applied_by":{"_account_id":28619,"name":"Dmitriy Rabotyagov","email":"noonedeadpunk@gmail.com","username":"noonedeadpunk"}},{"label":"Workflow","status":"MAY","applied_by":{"_account_id":25023,"name":"Jonathan Rosser","email":"jonathan.rosser@rd.bbc.co.uk","username":"jrosser"}},{"label":"Backport-Candidate","status":"MAY"}]}],"submit_requirements":[{"name":"Verified","description":"Verified in gate by CI","status":"SATISFIED","is_legacy":false,"submittability_expression_result":{"expression":"label:Verified\u003dMAX AND -label:Verified\u003dMIN","fulfilled":true,"status":"PASS","passing_atoms":["label:Verified\u003dMAX"],"failing_atoms":["label:Verified\u003dMIN"],"atom_explanations":{}}},{"name":"Backport-Candidate","description":"Backport candidate status","status":"NOT_APPLICABLE","is_legacy":false,"applicability_expression_result":{"fulfilled":false,"status":"FAIL"},"submittability_expression_result":{"expression":"is:true","fulfilled":true,"status":"NOT_EVALUATED","passing_atoms":[],"failing_atoms":[],"atom_explanations":{}}},{"name":"Code-Review","description":"Code reviewed by core reviewer","status":"SATISFIED","is_legacy":false,"submittability_expression_result":{"expression":"label:Code-Review\u003dMAX AND -label:Code-Review\u003dMIN","fulfilled":true,"status":"PASS","passing_atoms":["label:Code-Review\u003dMAX"],"failing_atoms":["label:Code-Review\u003dMIN"],"atom_explanations":{}}},{"name":"Workflow","description":"Approved for gate by core reviewer","status":"SATISFIED","is_legacy":false,"submittability_expression_result":{"expression":"label:Workflow\u003dMAX AND -label:Workflow\u003dMIN","fulfilled":true,"status":"PASS","passing_atoms":["label:Workflow\u003dMAX"],"failing_atoms":["label:Workflow\u003dMIN"],"atom_explanations":{}}}]}
