)]}'
{"id":"openstack%2Fopenstack-ansible-lxc_hosts~951932","triplet_id":"openstack%2Fopenstack-ansible-lxc_hosts~stable%2F2025.1~I60f1b04c2378c91905440e8de223c29dd6da4d17","project":"openstack/openstack-ansible-lxc_hosts","branch":"stable/2025.1","attention_set":{},"removed_from_attention_set":{"28619":{"account":{"_account_id":28619,"name":"Dmitriy Rabotyagov","email":"noonedeadpunk@gmail.com","username":"noonedeadpunk"},"last_update":"2025-06-07 15:01:43.000000000","reason":"Change was submitted"}},"hashtags":[],"change_id":"I60f1b04c2378c91905440e8de223c29dd6da4d17","subject":"Do not install COPR repo for CentOS LXC","status":"MERGED","created":"2025-06-06 07:05:46.000000000","updated":"2025-06-07 15:02:30.000000000","submitted":"2025-06-07 15:01:43.000000000","submitter":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"total_comment_count":1,"unresolved_comment_count":1,"has_review_started":true,"submission_id":"951932","meta_rev_id":"3c217fc30912ef54d8e4dd572640021a4ddfaa01","_number":951932,"virtual_id_number":951932,"owner":{"_account_id":28619,"name":"Dmitriy Rabotyagov","email":"noonedeadpunk@gmail.com","username":"noonedeadpunk"},"actions":{},"labels":{"Verified":{"approved":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"all":[{"value":0,"_account_id":34411,"name":"Neil Hanlon","email":"neil@shrug.pw","username":"nhanlon"},{"tag":"autogenerated:zuul:gate","value":2,"date":"2025-06-07 15:01:42.000000000","permitted_voting_range":{"min":2,"max":2},"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},{"value":0,"_account_id":25023,"name":"Jonathan Rosser","email":"jonathan.rosser@rd.bbc.co.uk","username":"jrosser"}],"values":{"-2":"Fails","-1":"Doesn\u0027t seem to work"," 0":"No score","+1":"Works for me","+2":"Verified"},"description":"","default_value":0,"optional":true},"Code-Review":{"approved":{"_account_id":34411,"name":"Neil Hanlon","email":"neil@shrug.pw","username":"nhanlon"},"all":[{"value":2,"date":"2025-06-07 13:05:00.000000000","permitted_voting_range":{"min":2,"max":2},"_account_id":34411,"name":"Neil Hanlon","email":"neil@shrug.pw","username":"nhanlon"},{"value":0,"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},{"value":2,"date":"2025-06-07 10:52:16.000000000","permitted_voting_range":{"min":2,"max":2},"_account_id":25023,"name":"Jonathan Rosser","email":"jonathan.rosser@rd.bbc.co.uk","username":"jrosser"}],"values":{"-2":"Do not merge","-1":"This patch needs further work before it can be merged"," 0":"No score","+1":"Looks good to me, but someone else must approve","+2":"Looks good to me (core reviewer)"},"description":"","default_value":0,"optional":true},"Workflow":{"approved":{"_account_id":34411,"name":"Neil Hanlon","email":"neil@shrug.pw","username":"nhanlon"},"all":[{"value":1,"date":"2025-06-07 13:05:00.000000000","permitted_voting_range":{"min":1,"max":1},"_account_id":34411,"name":"Neil Hanlon","email":"neil@shrug.pw","username":"nhanlon"},{"value":0,"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},{"value":0,"_account_id":25023,"name":"Jonathan Rosser","email":"jonathan.rosser@rd.bbc.co.uk","username":"jrosser"}],"values":{"-1":"Work in progress"," 0":"Ready for reviews","+1":"Approved"},"description":"","default_value":0,"optional":true},"Backport-Candidate":{"all":[{"value":0,"_account_id":34411,"name":"Neil Hanlon","email":"neil@shrug.pw","username":"nhanlon"},{"value":0,"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},{"value":0,"_account_id":25023,"name":"Jonathan Rosser","email":"jonathan.rosser@rd.bbc.co.uk","username":"jrosser"}],"values":{"-1":"Not A Backport Candidate"," 0":"Backport Review Needed","+1":"Proposed Backport"},"description":"","default_value":0,"optional":true}},"removable_reviewers":[],"reviewers":{"REVIEWER":[{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},{"_account_id":25023,"name":"Jonathan Rosser","email":"jonathan.rosser@rd.bbc.co.uk","username":"jrosser"},{"_account_id":34411,"name":"Neil Hanlon","email":"neil@shrug.pw","username":"nhanlon"}]},"pending_reviewers":{},"reviewer_updates":[{"updated":"2025-06-06 08:46:43.000000000","updated_by":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"reviewer":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"state":"REVIEWER"},{"updated":"2025-06-07 10:52:16.000000000","updated_by":{"_account_id":25023,"name":"Jonathan Rosser","email":"jonathan.rosser@rd.bbc.co.uk","username":"jrosser"},"reviewer":{"_account_id":25023,"name":"Jonathan Rosser","email":"jonathan.rosser@rd.bbc.co.uk","username":"jrosser"},"state":"REVIEWER"},{"updated":"2025-06-07 13:05:00.000000000","updated_by":{"_account_id":34411,"name":"Neil Hanlon","email":"neil@shrug.pw","username":"nhanlon"},"reviewer":{"_account_id":34411,"name":"Neil Hanlon","email":"neil@shrug.pw","username":"nhanlon"},"state":"REVIEWER"}],"messages":[{"id":"d167d5a4f24024ebaee384657dbce025c610fb23","tag":"autogenerated:gerrit:newPatchSet","author":{"_account_id":28619,"name":"Dmitriy Rabotyagov","email":"noonedeadpunk@gmail.com","username":"noonedeadpunk"},"date":"2025-06-06 07:05:46.000000000","message":"Uploaded patch set 1.","accounts_in_message":[],"_revision_number":1},{"id":"35d30c7caaf4c174bf3980f0ae1f44701f42df82","tag":"autogenerated:zuul:check","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2025-06-06 08:46:43.000000000","message":"Patch Set 1: Verified-1\n\nBuild failed (check pipeline).  For information on how to proceed, see\nhttps://docs.opendev.org/opendev/infra-manual/latest/developers.html#automated-testing\nand https://docs.openstack.org/project-team-guide/testing.html#how-to-handle-test-failures\n\nhttps://zuul.opendev.org/t/openstack/buildset/c07a5e07635846d0b2292f8c698b853a\n\n- openstack-ansible-linters-ubuntu-noble https://zuul.opendev.org/t/openstack/build/a23315d6cd3b432093aec53c569e1950 : SUCCESS in 5m 23s\n- openstack-tox-docs https://zuul.opendev.org/t/openstack/build/e99d999a70d5453ca0722cf8f890c64d : SUCCESS in 8m 10s\n- build-openstack-releasenotes https://zuul.opendev.org/t/openstack/build/314030c5ed2846929dcbc9618b1b15be : SUCCESS in 1m 59s\n- openstack-ansible-deploy-infra_lxc-centos-9-stream https://zuul.opendev.org/t/openstack/build/f2715aa0047a436f827c477644ab47df : FAILURE in 19m 09s (non-voting)\n- openstack-ansible-deploy-infra_lxc-rockylinux-9 https://zuul.opendev.org/t/openstack/build/be0a0b6b9a9844e681fcb216da03c464 : FAILURE in 15m 20s\n- openstack-ansible-deploy-infra_lxc_tls-rockylinux-9 https://zuul.opendev.org/t/openstack/build/838a38dcd260453a9b02af4425509292 : FAILURE in 13m 39s\n- openstack-ansible-deploy-infra_lxc-debian-bookworm https://zuul.opendev.org/t/openstack/build/5c3449e1e53647c59b8661a7a4b63788 : SUCCESS in 1h 22m 26s\n- openstack-ansible-deploy-infra_lxc-ubuntu-jammy https://zuul.opendev.org/t/openstack/build/a92d799def7140bfa60b5af45dd6a4ee : SUCCESS in 1h 16m 20s\n- openstack-ansible-deploy-infra_lxc_stepca-ubuntu-jammy https://zuul.opendev.org/t/openstack/build/5141649495384f7e9983b901818f4d6d : SUCCESS in 1h 13m 25s\n- openstack-ansible-deploy-infra_lxc_tls-ubuntu-jammy https://zuul.opendev.org/t/openstack/build/30162947987b4930bdbd8a7008610d71 : SUCCESS in 1h 32m 07s\n- openstack-ansible-deploy-infra_lxc-ubuntu-noble https://zuul.opendev.org/t/openstack/build/7bba39c1e8b146a4b64a08043258289d : SUCCESS in 1h 09m 35s\n- openstack-ansible-deploy-infra_lxc_stepca-ubuntu-noble https://zuul.opendev.org/t/openstack/build/25bbe7605a9a43888249ee06c80013b7 : SUCCESS in 1h 15m 07s\n- openstack-ansible-deploy-infra_lxc_tls-ubuntu-noble https://zuul.opendev.org/t/openstack/build/0e9351f186b540b9b5190c19dab61cb9 : SUCCESS in 1h 24m 54s\n- openstack-ansible-upgrade-infra_lxc-rockylinux-9 https://zuul.opendev.org/t/openstack/build/82a8fe23c3934127a3607938e7fa9798 : FAILURE in 1h 07m 48s\n- openstack-ansible-upgrade_2024.1-infra_lxc-rockylinux-9 https://zuul.opendev.org/t/openstack/build/bbdb50dc644a40f2be440afebbba7134 : FAILURE in 37m 04s\n- openstack-ansible-upgrade-infra_lxc-ubuntu-jammy https://zuul.opendev.org/t/openstack/build/a776fc6024b5468b97882f11f1aba36f : SUCCESS in 1h 28m 29s\n- openstack-ansible-upgrade_2024.1-infra_lxc-ubuntu-jammy https://zuul.opendev.org/t/openstack/build/542666750c7b481e89d0121e61e24b85 : SUCCESS in 1h 30m 37s\n- openstack-ansible-deploy-hosts_lxc_zfs-ubuntu-jammy https://zuul.opendev.org/t/openstack/build/479f7c573ac8421aa2471064b35b7906 : SUCCESS in 26m 10s\n- openstack-ansible-deploy-hosts_lxc_btrfs-ubuntu-jammy https://zuul.opendev.org/t/openstack/build/75cf06ed010249da917bda3ab4196217 : SUCCESS in 26m 01s\n- openstack-ansible-deploy-hosts_lxc_overlayfs-ubuntu-jammy https://zuul.opendev.org/t/openstack/build/2eee6469985042ff99bead2746d85c0f : SUCCESS in 12m 45s\n- openstack-ansible-deploy-hosts_lxc_zfs-ubuntu-noble https://zuul.opendev.org/t/openstack/build/7b54338073104683aa94e71cf954361f : SUCCESS in 13m 41s\n- openstack-ansible-deploy-hosts_lxc_btrfs-ubuntu-noble https://zuul.opendev.org/t/openstack/build/4a24eba21e5c4a7eb64d0c6157def09a : SUCCESS in 22m 02s\n- openstack-ansible-deploy-hosts_lxc_overlayfs-ubuntu-noble https://zuul.opendev.org/t/openstack/build/00b65072e7a849069f5490683668ecc8 : SUCCESS in 22m 40s\n- openstack-ansible-deploy-hosts_lxc_btrfs-debian-bookworm https://zuul.opendev.org/t/openstack/build/6d945777a3e34daa90c393232223488e : SUCCESS in 15m 58s\n- openstack-ansible-deploy-hosts_lxc_overlayfs-debian-bookworm https://zuul.opendev.org/t/openstack/build/52607509277d4556a94639f42becb571 : SUCCESS in 13m 58s\n- openstack-ansible-deploy-hosts_lxc_overlayfs-rockylinux-9 https://zuul.opendev.org/t/openstack/build/332f8ff7086e4c678c69410b73c53f6e : FAILURE in 12m 51s","accounts_in_message":[],"_revision_number":1},{"id":"429a77bdca022e7391a1b5fa204083287fb89a19","tag":"autogenerated:gerrit:newPatchSet","author":{"_account_id":28619,"name":"Dmitriy Rabotyagov","email":"noonedeadpunk@gmail.com","username":"noonedeadpunk"},"date":"2025-06-06 09:09:14.000000000","message":"Patch Set 2: Published edit on patch set 1.\n\nOutdated Votes:\n* Verified-1\n","accounts_in_message":[],"_revision_number":2},{"id":"14c5db1eea596d41357f224ca46effca19cdb54b","tag":"autogenerated:gerrit:newPatchSet","author":{"_account_id":28619,"name":"Dmitriy Rabotyagov","email":"noonedeadpunk@gmail.com","username":"noonedeadpunk"},"date":"2025-06-06 10:09:03.000000000","message":"Patch Set 3: Published edit on patch set 2.","accounts_in_message":[],"_revision_number":3},{"id":"993a5bcc6605c3c9697c85bfc4eae987fe4103d6","tag":"autogenerated:gerrit:newPatchSet","author":{"_account_id":28619,"name":"Dmitriy Rabotyagov","email":"noonedeadpunk@gmail.com","username":"noonedeadpunk"},"date":"2025-06-06 11:07:26.000000000","message":"Patch Set 4: Published edit on patch set 3.","accounts_in_message":[],"_revision_number":4},{"id":"a6c2f3d2f38d6d32d881de737b13b0b872450f4e","tag":"autogenerated:zuul:check","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2025-06-06 12:58:19.000000000","message":"Patch Set 4: Verified-1\n\nBuild failed (check pipeline).  For information on how to proceed, see\nhttps://docs.opendev.org/opendev/infra-manual/latest/developers.html#automated-testing\nand https://docs.openstack.org/project-team-guide/testing.html#how-to-handle-test-failures\n\nhttps://zuul.opendev.org/t/openstack/buildset/b0ad38e4d3084e2996c6eb07e1ba447f\n\n- openstack-ansible-linters-ubuntu-noble https://zuul.opendev.org/t/openstack/build/d2944ab52558463980f7045f0bc2d283 : SUCCESS in 7m 44s\n- openstack-tox-docs https://zuul.opendev.org/t/openstack/build/bb8d7f41c39543a39beefccc03213308 : SUCCESS in 6m 15s\n- build-openstack-releasenotes https://zuul.opendev.org/t/openstack/build/00d3af22249d413f885fdbddac4a466e : SUCCESS in 2m 35s\n- openstack-ansible-deploy-infra_lxc-centos-9-stream https://zuul.opendev.org/t/openstack/build/6f0941d885e34a4c8ca61cf391e26d2d : SUCCESS in 1h 14m 56s (non-voting)\n- openstack-ansible-deploy-infra_lxc-rockylinux-9 https://zuul.opendev.org/t/openstack/build/d24d464872d4422aafb2e4dce2adb365 : SUCCESS in 1h 19m 19s\n- openstack-ansible-deploy-infra_lxc_tls-rockylinux-9 https://zuul.opendev.org/t/openstack/build/7b3ac86fdf794fcdadb1118898a00327 : SUCCESS in 1h 17m 15s\n- openstack-ansible-deploy-infra_lxc-debian-bookworm https://zuul.opendev.org/t/openstack/build/a503294edbad4eb692d31f682c853259 : SUCCESS in 1h 12m 21s\n- openstack-ansible-deploy-infra_lxc-ubuntu-jammy https://zuul.opendev.org/t/openstack/build/ac5909d386574ff3889de8243e1ff2a9 : SUCCESS in 1h 12m 32s\n- openstack-ansible-deploy-infra_lxc_stepca-ubuntu-jammy https://zuul.opendev.org/t/openstack/build/c69138dfffe246e1b227725ef084daa9 : SUCCESS in 1h 19m 30s\n- openstack-ansible-deploy-infra_lxc_tls-ubuntu-jammy https://zuul.opendev.org/t/openstack/build/1aced521e1dc4ab583bf653db92c9090 : SUCCESS in 41m 25s\n- openstack-ansible-deploy-infra_lxc-ubuntu-noble https://zuul.opendev.org/t/openstack/build/87976e1b3fdf48d3bcbfb41d673916c6 : SUCCESS in 1h 30m 28s\n- openstack-ansible-deploy-infra_lxc_stepca-ubuntu-noble https://zuul.opendev.org/t/openstack/build/efc4ff24ccbc49d0bac8393382df08dd : SUCCESS in 50m 08s\n- openstack-ansible-deploy-infra_lxc_tls-ubuntu-noble https://zuul.opendev.org/t/openstack/build/f7337a3024334786a0770ae8106f9402 : SUCCESS in 40m 45s\n- openstack-ansible-upgrade-infra_lxc-rockylinux-9 https://zuul.opendev.org/t/openstack/build/6d92da22e2f2460bb1c09b2c651b7da7 : FAILURE in 1h 07m 48s\n- openstack-ansible-upgrade_2024.1-infra_lxc-rockylinux-9 https://zuul.opendev.org/t/openstack/build/39c61cb48203475ea6da3eb2c8d41165 : FAILURE in 42m 34s\n- openstack-ansible-upgrade-infra_lxc-ubuntu-jammy https://zuul.opendev.org/t/openstack/build/24fba578b92448ef9732e569eca7bdd8 : SUCCESS in 1h 42m 37s\n- openstack-ansible-upgrade_2024.1-infra_lxc-ubuntu-jammy https://zuul.opendev.org/t/openstack/build/039ef35b8f31405a8eebdc52c24ebb3b : SUCCESS in 52m 41s\n- openstack-ansible-deploy-hosts_lxc_zfs-ubuntu-jammy https://zuul.opendev.org/t/openstack/build/0293c4c75f5f4b58ba49a05fa37dc225 : SUCCESS in 15m 27s\n- openstack-ansible-deploy-hosts_lxc_btrfs-ubuntu-jammy https://zuul.opendev.org/t/openstack/build/54657f7e220b483890f91caba6334494 : SUCCESS in 21m 11s\n- openstack-ansible-deploy-hosts_lxc_overlayfs-ubuntu-jammy https://zuul.opendev.org/t/openstack/build/9b46bc56196442a688ee1f62814e5cbe : SUCCESS in 15m 05s\n- openstack-ansible-deploy-hosts_lxc_zfs-ubuntu-noble https://zuul.opendev.org/t/openstack/build/f492be5a9de34ac1840e2eca6ffe34ce : SUCCESS in 14m 53s\n- openstack-ansible-deploy-hosts_lxc_btrfs-ubuntu-noble https://zuul.opendev.org/t/openstack/build/07a0e2e7a5a24b90af4e8ab8494913a9 : SUCCESS in 24m 13s\n- openstack-ansible-deploy-hosts_lxc_overlayfs-ubuntu-noble https://zuul.opendev.org/t/openstack/build/6dbb9a1d9a2a46719ccce4781ef4a852 : SUCCESS in 23m 34s\n- openstack-ansible-deploy-hosts_lxc_btrfs-debian-bookworm https://zuul.opendev.org/t/openstack/build/2f5171b1cada453a8380f8df879d39d5 : SUCCESS in 21m 25s\n- openstack-ansible-deploy-hosts_lxc_overlayfs-debian-bookworm https://zuul.opendev.org/t/openstack/build/2e647b3ce74b4723b175e7d870925f5a : SUCCESS in 24m 12s\n- openstack-ansible-deploy-hosts_lxc_overlayfs-rockylinux-9 https://zuul.opendev.org/t/openstack/build/a166350702cd470a8499c461af8e1d16 : SUCCESS in 22m 53s","accounts_in_message":[],"_revision_number":4},{"id":"4b3ca1a0b4b8258bdaf00460962af4920349183b","tag":"autogenerated:gerrit:newPatchSet","author":{"_account_id":28619,"name":"Dmitriy Rabotyagov","email":"noonedeadpunk@gmail.com","username":"noonedeadpunk"},"date":"2025-06-06 13:13:27.000000000","message":"Patch Set 5: Published edit on patch set 4.\n\nOutdated Votes:\n* Verified-1\n","accounts_in_message":[],"_revision_number":5},{"id":"5231c025bbcb0099efa241237656fff013937f0a","tag":"autogenerated:gerrit:newPatchSet","author":{"_account_id":28619,"name":"Dmitriy Rabotyagov","email":"noonedeadpunk@gmail.com","username":"noonedeadpunk"},"date":"2025-06-06 13:15:42.000000000","message":"Patch Set 6: Published edit on patch set 5.","accounts_in_message":[],"_revision_number":6},{"id":"856b5732b9331a47a42aad5879f671269af1dd48","tag":"autogenerated:gerrit:newPatchSet","author":{"_account_id":28619,"name":"Dmitriy Rabotyagov","email":"noonedeadpunk@gmail.com","username":"noonedeadpunk"},"date":"2025-06-06 13:17:24.000000000","message":"Patch Set 7: Published edit on patch set 6.","accounts_in_message":[],"_revision_number":7},{"id":"3a37108481fb014147aba9ae5b1b6e4ab305e367","tag":"autogenerated:zuul:check","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2025-06-06 15:12:51.000000000","message":"Patch Set 7: Verified-1\n\nBuild failed (check pipeline).  For information on how to proceed, see\nhttps://docs.opendev.org/opendev/infra-manual/latest/developers.html#automated-testing\nand https://docs.openstack.org/project-team-guide/testing.html#how-to-handle-test-failures\n\nhttps://zuul.opendev.org/t/openstack/buildset/ccbd26dd426441db8dc7b6bd6c4eb89d\n\n- openstack-ansible-linters-ubuntu-noble https://zuul.opendev.org/t/openstack/build/5cff543460784ac2946e51fbd9544e67 : SUCCESS in 6m 52s\n- openstack-tox-docs https://zuul.opendev.org/t/openstack/build/3e8d0c1cd7224bdfac323d41e86f642e : SUCCESS in 5m 51s\n- openstack-ansible-deploy-infra_lxc-centos-9-stream https://zuul.opendev.org/t/openstack/build/934fc91621834eb699d37589b8369c55 : SUCCESS in 1h 14m 27s (non-voting)\n- openstack-ansible-deploy-infra_lxc-rockylinux-9 https://zuul.opendev.org/t/openstack/build/dde998f4880d4088a84f026ca34aba5c : SUCCESS in 1h 17m 07s\n- openstack-ansible-deploy-infra_lxc_tls-rockylinux-9 https://zuul.opendev.org/t/openstack/build/8dc132bc723143f79b8e12b37b088caf : SUCCESS in 1h 15m 12s\n- openstack-ansible-deploy-infra_lxc-debian-bookworm https://zuul.opendev.org/t/openstack/build/ee3094cff91a4e57b3e9d17873650af4 : SUCCESS in 39m 46s\n- openstack-ansible-deploy-infra_lxc-ubuntu-jammy https://zuul.opendev.org/t/openstack/build/fbf7881fbafd4811b1f57946ed0f492f : SUCCESS in 38m 47s\n- openstack-ansible-deploy-infra_lxc_stepca-ubuntu-jammy https://zuul.opendev.org/t/openstack/build/ff04cc7b46354f95bc78c7da0ff8fc52 : SUCCESS in 38m 58s\n- openstack-ansible-deploy-infra_lxc_tls-ubuntu-jammy https://zuul.opendev.org/t/openstack/build/4805071f8015459dac69a8ebef7a3401 : SUCCESS in 41m 21s\n- openstack-ansible-deploy-infra_lxc-ubuntu-noble https://zuul.opendev.org/t/openstack/build/00075a8360e64b0b8336906280c8c1b1 : SUCCESS in 50m 32s\n- openstack-ansible-deploy-infra_lxc_stepca-ubuntu-noble https://zuul.opendev.org/t/openstack/build/971da51b71764d01bd0ed76706e0f777 : SUCCESS in 44m 42s\n- openstack-ansible-deploy-infra_lxc_tls-ubuntu-noble https://zuul.opendev.org/t/openstack/build/53a463889cfd4858aafc4bc4df483ed1 : SUCCESS in 51m 47s\n- openstack-ansible-upgrade-infra_lxc-rockylinux-9 https://zuul.opendev.org/t/openstack/build/9409af4bc4324db5a34f19c790e3f264 : FAILURE in 35m 58s\n- openstack-ansible-upgrade_2024.1-infra_lxc-rockylinux-9 https://zuul.opendev.org/t/openstack/build/8f316bf1bb804fac8647ee903131f5e7 : FAILURE in 52m 43s\n- openstack-ansible-upgrade-infra_lxc-ubuntu-jammy https://zuul.opendev.org/t/openstack/build/712bdbb5f5cb4747a67c99f9ad047f50 : SUCCESS in 1h 39m 47s\n- openstack-ansible-upgrade_2024.1-infra_lxc-ubuntu-jammy https://zuul.opendev.org/t/openstack/build/4d48cb2b394449ada2278820e1b51dab : SUCCESS in 1h 47m 12s\n- openstack-ansible-deploy-hosts_lxc_zfs-ubuntu-jammy https://zuul.opendev.org/t/openstack/build/f062f2dafb6344919a87f58937c0e4d5 : SUCCESS in 21m 20s\n- openstack-ansible-deploy-hosts_lxc_btrfs-ubuntu-jammy https://zuul.opendev.org/t/openstack/build/301452a4c05c4908b4c15505b522a412 : SUCCESS in 15m 52s\n- openstack-ansible-deploy-hosts_lxc_overlayfs-ubuntu-jammy https://zuul.opendev.org/t/openstack/build/b94d460c2c764fd0a8cbd566b035121d : SUCCESS in 20m 30s\n- openstack-ansible-deploy-hosts_lxc_zfs-ubuntu-noble https://zuul.opendev.org/t/openstack/build/b23f30097e29402f9627f7c573542a4b : SUCCESS in 22m 50s\n- openstack-ansible-deploy-hosts_lxc_btrfs-ubuntu-noble https://zuul.opendev.org/t/openstack/build/39fe79aaaf1440f79a8e88b21a7c7648 : SUCCESS in 21m 41s\n- openstack-ansible-deploy-hosts_lxc_overlayfs-ubuntu-noble https://zuul.opendev.org/t/openstack/build/061df0cba5e34e0cae2e8341e344ad84 : SUCCESS in 18m 49s\n- openstack-ansible-deploy-hosts_lxc_btrfs-debian-bookworm https://zuul.opendev.org/t/openstack/build/70945fc9c8ca4bd6839390c164239673 : SUCCESS in 16m 25s\n- openstack-ansible-deploy-hosts_lxc_overlayfs-debian-bookworm https://zuul.opendev.org/t/openstack/build/538f44dea6c543f3a56ea14d6c323ba5 : SUCCESS in 21m 41s\n- openstack-ansible-deploy-hosts_lxc_overlayfs-rockylinux-9 https://zuul.opendev.org/t/openstack/build/2c6dd1043c82474c8b71aa0440e0b840 : SUCCESS in 24m 08s","accounts_in_message":[],"_revision_number":7},{"id":"9d8adf3ad2b0a49a54e39e8775e9f4278166bc92","tag":"autogenerated:gerrit:newPatchSet","author":{"_account_id":28619,"name":"Dmitriy Rabotyagov","email":"noonedeadpunk@gmail.com","username":"noonedeadpunk"},"date":"2025-06-06 18:59:55.000000000","message":"Patch Set 8: Published edit on patch set 7.\n\nOutdated Votes:\n* Verified-1\n","accounts_in_message":[],"_revision_number":8},{"id":"2b80ff0190bc03aa68623d169ab9c225429c2159","tag":"autogenerated:gerrit:newPatchSet","author":{"_account_id":28619,"name":"Dmitriy Rabotyagov","email":"noonedeadpunk@gmail.com","username":"noonedeadpunk"},"date":"2025-06-06 20:20:51.000000000","message":"Patch Set 9: Published edit on patch set 8.","accounts_in_message":[],"_revision_number":9},{"id":"ec6b97a8fd26f0db20c6f3a9651260d9d9884f15","tag":"autogenerated:zuul:check","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2025-06-06 22:28:27.000000000","message":"Patch Set 9: Verified+1\n\nBuild succeeded (check pipeline).\nhttps://zuul.opendev.org/t/openstack/buildset/19e42b275b42418f91e4c8a5030b8499\n\n- openstack-ansible-linters-ubuntu-noble https://zuul.opendev.org/t/openstack/build/7fd52df1be3e4843bad4b9e805a123b1 : SUCCESS in 5m 54s\n- openstack-tox-docs https://zuul.opendev.org/t/openstack/build/6ff3eeefc9b8425b83d5348157c6e537 : SUCCESS in 4m 48s\n- build-openstack-releasenotes https://zuul.opendev.org/t/openstack/build/4ad507f4881643b09abb07a37bc71d81 : SUCCESS in 2m 12s\n- openstack-ansible-deploy-infra_lxc-centos-9-stream https://zuul.opendev.org/t/openstack/build/7d4e6b1c94f642a08a7c16a111ceb1b2 : SUCCESS in 1h 02m 53s (non-voting)\n- openstack-ansible-deploy-infra_lxc-rockylinux-9 https://zuul.opendev.org/t/openstack/build/91e28ead672048889bc7c0ff93e28d6f : SUCCESS in 1h 21m 38s\n- openstack-ansible-deploy-infra_lxc_tls-rockylinux-9 https://zuul.opendev.org/t/openstack/build/900dd49b551742f78a3285f0f59752d1 : SUCCESS in 42m 59s\n- openstack-ansible-deploy-infra_lxc-debian-bookworm https://zuul.opendev.org/t/openstack/build/944acff6eaae45eabda419a02b598453 : SUCCESS in 41m 40s\n- openstack-ansible-deploy-infra_lxc-ubuntu-jammy https://zuul.opendev.org/t/openstack/build/e6a181b3ab7443fbaeee011f70e3380f : SUCCESS in 38m 41s\n- openstack-ansible-deploy-infra_lxc_stepca-ubuntu-jammy https://zuul.opendev.org/t/openstack/build/63d0d1bcd94048e49372f6205a33651b : SUCCESS in 1h 40m 11s\n- openstack-ansible-deploy-infra_lxc_tls-ubuntu-jammy https://zuul.opendev.org/t/openstack/build/48a5003dda0f4cec818ac96c0aea7e15 : SUCCESS in 1h 11m 58s\n- openstack-ansible-deploy-infra_lxc-ubuntu-noble https://zuul.opendev.org/t/openstack/build/f6c006f728be416aa522e6400f8dcafc : SUCCESS in 1h 13m 37s\n- openstack-ansible-deploy-infra_lxc_stepca-ubuntu-noble https://zuul.opendev.org/t/openstack/build/af399875937e4db79f0e2683c5ff3af4 : SUCCESS in 1h 20m 57s\n- openstack-ansible-deploy-infra_lxc_tls-ubuntu-noble https://zuul.opendev.org/t/openstack/build/da8115b8201d4344a98a30018cb19d29 : SUCCESS in 1h 29m 10s\n- openstack-ansible-upgrade-infra_lxc-rockylinux-9 https://zuul.opendev.org/t/openstack/build/d179b8fbbb6c450c90abc338efa1dbd1 : SUCCESS in 1h 40m 32s\n- openstack-ansible-upgrade_2024.1-infra_lxc-rockylinux-9 https://zuul.opendev.org/t/openstack/build/946f41e2312d4bd78a8d7329418a5657 : SUCCESS in 1h 44m 01s\n- openstack-ansible-upgrade-infra_lxc-ubuntu-jammy https://zuul.opendev.org/t/openstack/build/2d8d3711f99e476eb45aca05869779f9 : SUCCESS in 1h 32m 57s\n- openstack-ansible-upgrade_2024.1-infra_lxc-ubuntu-jammy https://zuul.opendev.org/t/openstack/build/8e9f2447c2bf4cf4ae4b0bf747b70e6a : SUCCESS in 2h 00m 15s\n- openstack-ansible-deploy-hosts_lxc_zfs-ubuntu-jammy https://zuul.opendev.org/t/openstack/build/203b5de251e944f295f9955fec6d6524 : SUCCESS in 21m 09s\n- openstack-ansible-deploy-hosts_lxc_btrfs-ubuntu-jammy https://zuul.opendev.org/t/openstack/build/258da16b987b45f688fff9f1c23a4ecd : SUCCESS in 24m 35s\n- openstack-ansible-deploy-hosts_lxc_overlayfs-ubuntu-jammy https://zuul.opendev.org/t/openstack/build/17361f2a6d6c4466980a74e2c030168c : SUCCESS in 24m 05s\n- openstack-ansible-deploy-hosts_lxc_zfs-ubuntu-noble https://zuul.opendev.org/t/openstack/build/0323165f3aba45119bd6bbe92159da3d : SUCCESS in 25m 36s\n- openstack-ansible-deploy-hosts_lxc_btrfs-ubuntu-noble https://zuul.opendev.org/t/openstack/build/3edbfdcb00d64aad8fa10a5540f24bcd : SUCCESS in 22m 05s\n- openstack-ansible-deploy-hosts_lxc_overlayfs-ubuntu-noble https://zuul.opendev.org/t/openstack/build/d8198315a08f47b5836bf14dd3a37402 : SUCCESS in 21m 39s\n- openstack-ansible-deploy-hosts_lxc_btrfs-debian-bookworm https://zuul.opendev.org/t/openstack/build/a922adff192b4acfaf4413b35a0e013c : SUCCESS in 25m 59s\n- openstack-ansible-deploy-hosts_lxc_overlayfs-debian-bookworm https://zuul.opendev.org/t/openstack/build/e0088c46e8c049be86ed98ac72621ef3 : SUCCESS in 19m 53s\n- openstack-ansible-deploy-hosts_lxc_overlayfs-rockylinux-9 https://zuul.opendev.org/t/openstack/build/1d279ec14c924768b8f65cea3f7712c7 : SUCCESS in 22m 51s","accounts_in_message":[],"_revision_number":9},{"id":"fcfe92aa000c650338b576b8baf811d4ca12a257","author":{"_account_id":28619,"name":"Dmitriy Rabotyagov","email":"noonedeadpunk@gmail.com","username":"noonedeadpunk"},"date":"2025-06-07 08:08:22.000000000","message":"Patch Set 9:\n\n(1 comment)","accounts_in_message":[],"_revision_number":9},{"id":"af51f9a856ea1a8247f45b8f12e60f68c903671d","author":{"_account_id":25023,"name":"Jonathan Rosser","email":"jonathan.rosser@rd.bbc.co.uk","username":"jrosser"},"date":"2025-06-07 10:52:16.000000000","message":"Patch Set 9: Code-Review+2","accounts_in_message":[],"_revision_number":9},{"id":"b7e615d1ef85ec0b8c56d6c640d0ec1448355f6a","author":{"_account_id":34411,"name":"Neil Hanlon","email":"neil@shrug.pw","username":"nhanlon"},"date":"2025-06-07 13:05:00.000000000","message":"Patch Set 9: Code-Review+2 Workflow+1","accounts_in_message":[],"_revision_number":9},{"id":"8b66e119c2863065183d622135972cef1e626796","tag":"autogenerated:zuul:gate","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2025-06-07 13:06:52.000000000","message":"Patch Set 9: -Verified\n\nStarting gate jobs.","accounts_in_message":[],"_revision_number":9},{"id":"073a7183cea7d9a115961a57b76f2faf35945f3b","tag":"autogenerated:zuul:gate","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2025-06-07 15:01:42.000000000","message":"Patch Set 9: Verified+2\n\nBuild succeeded (gate pipeline).\nhttps://zuul.opendev.org/t/openstack/buildset/a7da67c576e34113b9648b1208693cc8\n\n- openstack-ansible-linters-ubuntu-noble https://zuul.opendev.org/t/openstack/build/36a0b83182614d5bbff5cb0bb0af103c : SUCCESS in 6m 21s\n- openstack-tox-docs https://zuul.opendev.org/t/openstack/build/2a12b713896745d794cca457fb7a9fbb : SUCCESS in 5m 31s\n- build-openstack-releasenotes https://zuul.opendev.org/t/openstack/build/8fdb6308413b493798e019dd00398b3c : SUCCESS in 2m 17s\n- openstack-ansible-deploy-infra_lxc-rockylinux-9 https://zuul.opendev.org/t/openstack/build/c80cf22602ef4d9298ee2dd6525896c3 : SUCCESS in 44m 35s\n- openstack-ansible-deploy-infra_lxc_tls-rockylinux-9 https://zuul.opendev.org/t/openstack/build/218c1c7b73114e0da3d60fc40b961624 : SUCCESS in 43m 04s\n- openstack-ansible-deploy-infra_lxc-debian-bookworm https://zuul.opendev.org/t/openstack/build/4a9005a711734960a963d5f53fd1a3c5 : SUCCESS in 41m 40s\n- openstack-ansible-deploy-infra_lxc-ubuntu-jammy https://zuul.opendev.org/t/openstack/build/36d3e39b5fb04a6aaa25da3f12b4b16f : SUCCESS in 38m 49s\n- openstack-ansible-deploy-infra_lxc_stepca-ubuntu-jammy https://zuul.opendev.org/t/openstack/build/5cc14086fe424223854673d43f147076 : SUCCESS in 39m 06s\n- openstack-ansible-deploy-infra_lxc_tls-ubuntu-jammy https://zuul.opendev.org/t/openstack/build/0672c895d3cf40258a96a069d3fd4091 : SUCCESS in 44m 04s\n- openstack-ansible-deploy-infra_lxc-ubuntu-noble https://zuul.opendev.org/t/openstack/build/fa590b3810a94930b5a8ab7944fb8370 : SUCCESS in 1h 16m 22s\n- openstack-ansible-deploy-infra_lxc_stepca-ubuntu-noble https://zuul.opendev.org/t/openstack/build/f7a8cc3f30c94ef9906fcfe274e35bed : SUCCESS in 50m 02s\n- openstack-ansible-deploy-infra_lxc_tls-ubuntu-noble https://zuul.opendev.org/t/openstack/build/08e4d13a09c04ad18030544028d41992 : SUCCESS in 41m 04s\n- openstack-ansible-upgrade-infra_lxc-rockylinux-9 https://zuul.opendev.org/t/openstack/build/3532733cbc9440e88dbe9a58c1a160d5 : SUCCESS in 1h 50m 15s\n- openstack-ansible-upgrade_2024.1-infra_lxc-rockylinux-9 https://zuul.opendev.org/t/openstack/build/255df72e545841aa8dae50de353836a0 : SUCCESS in 1h 06m 35s\n- openstack-ansible-upgrade-infra_lxc-ubuntu-jammy https://zuul.opendev.org/t/openstack/build/cd11bcec187042ab9a1e5c2366934371 : SUCCESS in 47m 54s\n- openstack-ansible-upgrade_2024.1-infra_lxc-ubuntu-jammy https://zuul.opendev.org/t/openstack/build/6a0aaff5da0d437bbb135f461930045f : SUCCESS in 1h 21m 43s\n- openstack-ansible-deploy-hosts_lxc_zfs-ubuntu-jammy https://zuul.opendev.org/t/openstack/build/6300d20a15644e44925ee966ccedc36c : SUCCESS in 20m 39s\n- openstack-ansible-deploy-hosts_lxc_btrfs-ubuntu-jammy https://zuul.opendev.org/t/openstack/build/84c07c41824b409283dc66d0c55961ae : SUCCESS in 18m 06s\n- openstack-ansible-deploy-hosts_lxc_overlayfs-ubuntu-jammy https://zuul.opendev.org/t/openstack/build/294c273974cc4947bffd527199f059c6 : SUCCESS in 19m 05s\n- openstack-ansible-deploy-hosts_lxc_zfs-ubuntu-noble https://zuul.opendev.org/t/openstack/build/3198b75ec08d4849bde435102e09ea3c : SUCCESS in 20m 21s\n- openstack-ansible-deploy-hosts_lxc_btrfs-ubuntu-noble https://zuul.opendev.org/t/openstack/build/d39d22bca6c04d678354585eb0992388 : SUCCESS in 20m 58s\n- openstack-ansible-deploy-hosts_lxc_overlayfs-ubuntu-noble https://zuul.opendev.org/t/openstack/build/8de6c1af94ed4eef9c04d8141144a31f : SUCCESS in 19m 46s\n- openstack-ansible-deploy-hosts_lxc_btrfs-debian-bookworm https://zuul.opendev.org/t/openstack/build/be040523555240dc9b4dad3a06abe07d : SUCCESS in 21m 15s\n- openstack-ansible-deploy-hosts_lxc_overlayfs-debian-bookworm https://zuul.opendev.org/t/openstack/build/c01cb8c0a7704d8fa934bbe7f384a1f2 : SUCCESS in 22m 02s\n- openstack-ansible-deploy-hosts_lxc_overlayfs-rockylinux-9 https://zuul.opendev.org/t/openstack/build/a3d9aee67b9a49ee9033cc274d4761f3 : SUCCESS in 22m 31s","accounts_in_message":[],"_revision_number":9},{"id":"8c453a31bbd3b8a72d695e9e0fc6405bd9ace00c","tag":"autogenerated:gerrit:merged","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2025-06-07 15:01:43.000000000","message":"Change has been successfully merged","accounts_in_message":[],"_revision_number":9},{"id":"3c217fc30912ef54d8e4dd572640021a4ddfaa01","tag":"autogenerated:zuul:promote","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2025-06-07 15:02:30.000000000","message":"Patch Set 9:\n\nBuild succeeded (promote pipeline).\nhttps://zuul.opendev.org/t/openstack/buildset/73f4eebf70f54519b64e54c579d91d2a\n\n- promote-openstack-tox-docs https://zuul.opendev.org/t/openstack/build/3c4d56ed245c49eba5ddda8329df7dca : SUCCESS in 34s","accounts_in_message":[],"_revision_number":9}],"current_revision_number":9,"current_revision":"1de59fb340868a7fc6b734556a23c2d312dcc469","revisions":{"1362355d9d6e8e73c34f92eeb737fffbaec66d9e":{"kind":"REWORK","_number":1,"created":"2025-06-06 07:05:46.000000000","uploader":{"_account_id":28619,"name":"Dmitriy Rabotyagov","email":"noonedeadpunk@gmail.com","username":"noonedeadpunk"},"ref":"refs/changes/32/951932/1","fetch":{"anonymous http":{"url":"https://review.opendev.org/openstack/openstack-ansible-lxc_hosts","ref":"refs/changes/32/951932/1","commands":{"Checkout":"git fetch https://review.opendev.org/openstack/openstack-ansible-lxc_hosts refs/changes/32/951932/1 \u0026\u0026 git checkout FETCH_HEAD","Cherry Pick":"git fetch https://review.opendev.org/openstack/openstack-ansible-lxc_hosts refs/changes/32/951932/1 \u0026\u0026 git cherry-pick FETCH_HEAD","Format Patch":"git fetch https://review.opendev.org/openstack/openstack-ansible-lxc_hosts refs/changes/32/951932/1 \u0026\u0026 git format-patch -1 --stdout FETCH_HEAD","Pull":"git pull https://review.opendev.org/openstack/openstack-ansible-lxc_hosts refs/changes/32/951932/1"}}},"commit":{"parents":[{"commit":"b69a166731f7d23849100c6ad9ff999ddf0ca9d3","subject":"Do not use distro templates by default","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/openstack-ansible-lxc_hosts/commit/b69a166731f7d23849100c6ad9ff999ddf0ca9d3"}]}],"author":{"name":"Dmitriy Rabotyagov","email":"dmitriy.rabotyagov@citynetwork.eu","date":"2022-05-27 18:19:35.000000000","tz":120},"committer":{"name":"Dmitriy Rabotyagov","email":"dmitriy.rabotyagov@cleura.com","date":"2025-06-06 07:05:04.000000000","tz":120},"subject":"Do not install COPR repo for CentOS LXC","message":"Do not install COPR repo for CentOS LXC\n\nEPEL does contain LXC package for CentOS 8 and 9 Stream, so we can\ninstall LXC from there. For CentOS 9 COPR still needed for lxc-templates\n\nWith that we also add control of what packages are going to be installed\nfrom EPEL.\n\nChange-Id: I60f1b04c2378c91905440e8de223c29dd6da4d17\n(cherry picked from commit c84b23b25851a8fca67a5a6e70d8bc687d5832d5)\n","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/openstack-ansible-lxc_hosts/commit/1362355d9d6e8e73c34f92eeb737fffbaec66d9e"}],"resolve_conflicts_web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/openstack-ansible-lxc_hosts/commit/1362355d9d6e8e73c34f92eeb737fffbaec66d9e"}]},"branch":"refs/heads/stable/2025.1"},"59db151ab97fbf8e597e6a6b59cc988262c47f92":{"kind":"REWORK","_number":2,"created":"2025-06-06 09:09:14.000000000","uploader":{"_account_id":28619,"name":"Dmitriy Rabotyagov","email":"noonedeadpunk@gmail.com","username":"noonedeadpunk"},"ref":"refs/changes/32/951932/2","fetch":{"anonymous http":{"url":"https://review.opendev.org/openstack/openstack-ansible-lxc_hosts","ref":"refs/changes/32/951932/2","commands":{"Checkout":"git fetch https://review.opendev.org/openstack/openstack-ansible-lxc_hosts refs/changes/32/951932/2 \u0026\u0026 git checkout FETCH_HEAD","Cherry Pick":"git fetch https://review.opendev.org/openstack/openstack-ansible-lxc_hosts refs/changes/32/951932/2 \u0026\u0026 git cherry-pick FETCH_HEAD","Format Patch":"git fetch https://review.opendev.org/openstack/openstack-ansible-lxc_hosts refs/changes/32/951932/2 \u0026\u0026 git format-patch -1 --stdout FETCH_HEAD","Pull":"git pull https://review.opendev.org/openstack/openstack-ansible-lxc_hosts refs/changes/32/951932/2"}}},"commit":{"parents":[{"commit":"b69a166731f7d23849100c6ad9ff999ddf0ca9d3","subject":"Do not use distro templates by default","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/openstack-ansible-lxc_hosts/commit/b69a166731f7d23849100c6ad9ff999ddf0ca9d3"}]}],"author":{"name":"Dmitriy Rabotyagov","email":"dmitriy.rabotyagov@citynetwork.eu","date":"2022-05-27 18:19:35.000000000","tz":120},"committer":{"name":"Dmitriy Rabotyagov","email":"dmitriy.rabotyagov@cleura.com","date":"2025-06-06 09:09:14.000000000","tz":0},"subject":"Do not install COPR repo for CentOS LXC","message":"Do not install COPR repo for CentOS LXC\n\nEPEL does contain LXC package for CentOS 8 and 9 Stream, so we can\ninstall LXC from there. For CentOS 9 COPR still needed for lxc-templates\n\nWith that we also add control of what packages are going to be installed\nfrom EPEL.\n\nChange-Id: I60f1b04c2378c91905440e8de223c29dd6da4d17\n(cherry picked from commit c84b23b25851a8fca67a5a6e70d8bc687d5832d5)\n","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/openstack-ansible-lxc_hosts/commit/59db151ab97fbf8e597e6a6b59cc988262c47f92"}],"resolve_conflicts_web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/openstack-ansible-lxc_hosts/commit/59db151ab97fbf8e597e6a6b59cc988262c47f92"}]},"branch":"refs/heads/stable/2025.1"},"40476ed6ed6e170e73de01dbf8e62ea1e714503e":{"kind":"REWORK","_number":3,"created":"2025-06-06 10:09:03.000000000","uploader":{"_account_id":28619,"name":"Dmitriy Rabotyagov","email":"noonedeadpunk@gmail.com","username":"noonedeadpunk"},"ref":"refs/changes/32/951932/3","fetch":{"anonymous http":{"url":"https://review.opendev.org/openstack/openstack-ansible-lxc_hosts","ref":"refs/changes/32/951932/3","commands":{"Checkout":"git fetch https://review.opendev.org/openstack/openstack-ansible-lxc_hosts refs/changes/32/951932/3 \u0026\u0026 git checkout FETCH_HEAD","Cherry Pick":"git fetch https://review.opendev.org/openstack/openstack-ansible-lxc_hosts refs/changes/32/951932/3 \u0026\u0026 git cherry-pick FETCH_HEAD","Format Patch":"git fetch https://review.opendev.org/openstack/openstack-ansible-lxc_hosts refs/changes/32/951932/3 \u0026\u0026 git format-patch -1 --stdout FETCH_HEAD","Pull":"git pull https://review.opendev.org/openstack/openstack-ansible-lxc_hosts refs/changes/32/951932/3"}}},"commit":{"parents":[{"commit":"b69a166731f7d23849100c6ad9ff999ddf0ca9d3","subject":"Do not use distro templates by default","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/openstack-ansible-lxc_hosts/commit/b69a166731f7d23849100c6ad9ff999ddf0ca9d3"}]}],"author":{"name":"Dmitriy Rabotyagov","email":"dmitriy.rabotyagov@citynetwork.eu","date":"2022-05-27 18:19:35.000000000","tz":120},"committer":{"name":"Dmitriy Rabotyagov","email":"dmitriy.rabotyagov@cleura.com","date":"2025-06-06 10:09:03.000000000","tz":0},"subject":"Do not install COPR repo for CentOS LXC","message":"Do not install COPR repo for CentOS LXC\n\nEPEL does contain LXC package for CentOS 8 and 9 Stream, so we can\ninstall LXC from there. For CentOS 9 COPR still needed for lxc-templates\n\nWith that we also add control of what packages are going to be installed\nfrom EPEL.\n\nChange-Id: I60f1b04c2378c91905440e8de223c29dd6da4d17\n(cherry picked from commit c84b23b25851a8fca67a5a6e70d8bc687d5832d5)\n","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/openstack-ansible-lxc_hosts/commit/40476ed6ed6e170e73de01dbf8e62ea1e714503e"}],"resolve_conflicts_web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/openstack-ansible-lxc_hosts/commit/40476ed6ed6e170e73de01dbf8e62ea1e714503e"}]},"branch":"refs/heads/stable/2025.1"},"f41915c0606a29f6443fd280f981d1f975e60023":{"kind":"REWORK","_number":4,"created":"2025-06-06 11:07:26.000000000","uploader":{"_account_id":28619,"name":"Dmitriy Rabotyagov","email":"noonedeadpunk@gmail.com","username":"noonedeadpunk"},"ref":"refs/changes/32/951932/4","fetch":{"anonymous http":{"url":"https://review.opendev.org/openstack/openstack-ansible-lxc_hosts","ref":"refs/changes/32/951932/4","commands":{"Checkout":"git fetch https://review.opendev.org/openstack/openstack-ansible-lxc_hosts refs/changes/32/951932/4 \u0026\u0026 git checkout FETCH_HEAD","Cherry Pick":"git fetch https://review.opendev.org/openstack/openstack-ansible-lxc_hosts refs/changes/32/951932/4 \u0026\u0026 git cherry-pick FETCH_HEAD","Format Patch":"git fetch https://review.opendev.org/openstack/openstack-ansible-lxc_hosts refs/changes/32/951932/4 \u0026\u0026 git format-patch -1 --stdout FETCH_HEAD","Pull":"git pull https://review.opendev.org/openstack/openstack-ansible-lxc_hosts refs/changes/32/951932/4"}}},"commit":{"parents":[{"commit":"b69a166731f7d23849100c6ad9ff999ddf0ca9d3","subject":"Do not use distro templates by default","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/openstack-ansible-lxc_hosts/commit/b69a166731f7d23849100c6ad9ff999ddf0ca9d3"}]}],"author":{"name":"Dmitriy Rabotyagov","email":"dmitriy.rabotyagov@citynetwork.eu","date":"2022-05-27 18:19:35.000000000","tz":120},"committer":{"name":"Dmitriy Rabotyagov","email":"dmitriy.rabotyagov@cleura.com","date":"2025-06-06 11:07:26.000000000","tz":0},"subject":"Do not install COPR repo for CentOS LXC","message":"Do not install COPR repo for CentOS LXC\n\nEPEL does contain LXC package for CentOS 8 and 9 Stream, so we can\ninstall LXC from there. For CentOS 9 COPR still needed for lxc-templates\n\nWith that we also add control of what packages are going to be installed\nfrom EPEL.\n\nChange-Id: I60f1b04c2378c91905440e8de223c29dd6da4d17\n(cherry picked from commit c84b23b25851a8fca67a5a6e70d8bc687d5832d5)\n","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/openstack-ansible-lxc_hosts/commit/f41915c0606a29f6443fd280f981d1f975e60023"}],"resolve_conflicts_web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/openstack-ansible-lxc_hosts/commit/f41915c0606a29f6443fd280f981d1f975e60023"}]},"branch":"refs/heads/stable/2025.1"},"8a3f5ee363b71f5e9fe0d1bbb7bfbc89e7c3cc42":{"kind":"REWORK","_number":5,"created":"2025-06-06 13:13:27.000000000","uploader":{"_account_id":28619,"name":"Dmitriy Rabotyagov","email":"noonedeadpunk@gmail.com","username":"noonedeadpunk"},"ref":"refs/changes/32/951932/5","fetch":{"anonymous http":{"url":"https://review.opendev.org/openstack/openstack-ansible-lxc_hosts","ref":"refs/changes/32/951932/5","commands":{"Checkout":"git fetch https://review.opendev.org/openstack/openstack-ansible-lxc_hosts refs/changes/32/951932/5 \u0026\u0026 git checkout FETCH_HEAD","Cherry Pick":"git fetch https://review.opendev.org/openstack/openstack-ansible-lxc_hosts refs/changes/32/951932/5 \u0026\u0026 git cherry-pick FETCH_HEAD","Format Patch":"git fetch https://review.opendev.org/openstack/openstack-ansible-lxc_hosts refs/changes/32/951932/5 \u0026\u0026 git format-patch -1 --stdout FETCH_HEAD","Pull":"git pull https://review.opendev.org/openstack/openstack-ansible-lxc_hosts refs/changes/32/951932/5"}}},"commit":{"parents":[{"commit":"b69a166731f7d23849100c6ad9ff999ddf0ca9d3","subject":"Do not use distro templates by default","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/openstack-ansible-lxc_hosts/commit/b69a166731f7d23849100c6ad9ff999ddf0ca9d3"}]}],"author":{"name":"Dmitriy Rabotyagov","email":"dmitriy.rabotyagov@citynetwork.eu","date":"2022-05-27 18:19:35.000000000","tz":120},"committer":{"name":"Dmitriy Rabotyagov","email":"dmitriy.rabotyagov@cleura.com","date":"2025-06-06 13:13:27.000000000","tz":0},"subject":"Do not install COPR repo for CentOS LXC","message":"Do not install COPR repo for CentOS LXC\n\nEPEL does contain LXC package for CentOS 8 and 9 Stream, so we can\ninstall LXC from there. For CentOS 9 COPR still needed for lxc-templates\n\nWith that we also add control of what packages are going to be installed\nfrom EPEL.\n\nChange-Id: I60f1b04c2378c91905440e8de223c29dd6da4d17\n(cherry picked from commit c84b23b25851a8fca67a5a6e70d8bc687d5832d5)\n","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/openstack-ansible-lxc_hosts/commit/8a3f5ee363b71f5e9fe0d1bbb7bfbc89e7c3cc42"}],"resolve_conflicts_web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/openstack-ansible-lxc_hosts/commit/8a3f5ee363b71f5e9fe0d1bbb7bfbc89e7c3cc42"}]},"branch":"refs/heads/stable/2025.1"},"95fb2ee5853ab1c8de93e56162fa3598873cd2b1":{"kind":"REWORK","_number":6,"created":"2025-06-06 13:15:42.000000000","uploader":{"_account_id":28619,"name":"Dmitriy Rabotyagov","email":"noonedeadpunk@gmail.com","username":"noonedeadpunk"},"ref":"refs/changes/32/951932/6","fetch":{"anonymous http":{"url":"https://review.opendev.org/openstack/openstack-ansible-lxc_hosts","ref":"refs/changes/32/951932/6","commands":{"Checkout":"git fetch https://review.opendev.org/openstack/openstack-ansible-lxc_hosts refs/changes/32/951932/6 \u0026\u0026 git checkout FETCH_HEAD","Cherry Pick":"git fetch https://review.opendev.org/openstack/openstack-ansible-lxc_hosts refs/changes/32/951932/6 \u0026\u0026 git cherry-pick FETCH_HEAD","Format Patch":"git fetch https://review.opendev.org/openstack/openstack-ansible-lxc_hosts refs/changes/32/951932/6 \u0026\u0026 git format-patch -1 --stdout FETCH_HEAD","Pull":"git pull https://review.opendev.org/openstack/openstack-ansible-lxc_hosts refs/changes/32/951932/6"}}},"commit":{"parents":[{"commit":"b69a166731f7d23849100c6ad9ff999ddf0ca9d3","subject":"Do not use distro templates by default","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/openstack-ansible-lxc_hosts/commit/b69a166731f7d23849100c6ad9ff999ddf0ca9d3"}]}],"author":{"name":"Dmitriy Rabotyagov","email":"dmitriy.rabotyagov@citynetwork.eu","date":"2022-05-27 18:19:35.000000000","tz":120},"committer":{"name":"Dmitriy Rabotyagov","email":"dmitriy.rabotyagov@cleura.com","date":"2025-06-06 13:15:42.000000000","tz":0},"subject":"Do not install COPR repo for CentOS LXC","message":"Do not install COPR repo for CentOS LXC\n\nEPEL does contain LXC package for CentOS 8 and 9 Stream, so we can\ninstall LXC from there. For CentOS 9 COPR still needed for lxc-templates\n\nWith that we also add control of what packages are going to be installed\nfrom EPEL.\n\nChange-Id: I60f1b04c2378c91905440e8de223c29dd6da4d17\n(cherry picked from commit c84b23b25851a8fca67a5a6e70d8bc687d5832d5)\n","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/openstack-ansible-lxc_hosts/commit/95fb2ee5853ab1c8de93e56162fa3598873cd2b1"}],"resolve_conflicts_web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/openstack-ansible-lxc_hosts/commit/95fb2ee5853ab1c8de93e56162fa3598873cd2b1"}]},"branch":"refs/heads/stable/2025.1"},"4eaa45ba029ed383ca77e0a7e5fb3f4312721e70":{"kind":"REWORK","_number":7,"created":"2025-06-06 13:17:24.000000000","uploader":{"_account_id":28619,"name":"Dmitriy Rabotyagov","email":"noonedeadpunk@gmail.com","username":"noonedeadpunk"},"ref":"refs/changes/32/951932/7","fetch":{"anonymous http":{"url":"https://review.opendev.org/openstack/openstack-ansible-lxc_hosts","ref":"refs/changes/32/951932/7","commands":{"Checkout":"git fetch https://review.opendev.org/openstack/openstack-ansible-lxc_hosts refs/changes/32/951932/7 \u0026\u0026 git checkout FETCH_HEAD","Cherry Pick":"git fetch https://review.opendev.org/openstack/openstack-ansible-lxc_hosts refs/changes/32/951932/7 \u0026\u0026 git cherry-pick FETCH_HEAD","Format Patch":"git fetch https://review.opendev.org/openstack/openstack-ansible-lxc_hosts refs/changes/32/951932/7 \u0026\u0026 git format-patch -1 --stdout FETCH_HEAD","Pull":"git pull https://review.opendev.org/openstack/openstack-ansible-lxc_hosts refs/changes/32/951932/7"}}},"commit":{"parents":[{"commit":"b69a166731f7d23849100c6ad9ff999ddf0ca9d3","subject":"Do not use distro templates by default","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/openstack-ansible-lxc_hosts/commit/b69a166731f7d23849100c6ad9ff999ddf0ca9d3"}]}],"author":{"name":"Dmitriy Rabotyagov","email":"dmitriy.rabotyagov@citynetwork.eu","date":"2022-05-27 18:19:35.000000000","tz":120},"committer":{"name":"Dmitriy Rabotyagov","email":"dmitriy.rabotyagov@cleura.com","date":"2025-06-06 13:17:01.000000000","tz":0},"subject":"Do not install COPR repo for CentOS LXC","message":"Do not install COPR repo for CentOS LXC\n\nEPEL does contain LXC package for CentOS 8 and 9 Stream, so we can\ninstall LXC from there. For CentOS 9 COPR still needed for lxc-templates\n\nWith that we also add control of what packages are going to be installed\nfrom EPEL.\n\nChange-Id: I60f1b04c2378c91905440e8de223c29dd6da4d17\n(cherry picked from commit c84b23b25851a8fca67a5a6e70d8bc687d5832d5)\n","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/openstack-ansible-lxc_hosts/commit/4eaa45ba029ed383ca77e0a7e5fb3f4312721e70"}],"resolve_conflicts_web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/openstack-ansible-lxc_hosts/commit/4eaa45ba029ed383ca77e0a7e5fb3f4312721e70"}]},"branch":"refs/heads/stable/2025.1"},"bca3f2b11a73ae1e82c93eeabf06f1b11179d475":{"kind":"REWORK","_number":8,"created":"2025-06-06 18:59:55.000000000","uploader":{"_account_id":28619,"name":"Dmitriy Rabotyagov","email":"noonedeadpunk@gmail.com","username":"noonedeadpunk"},"ref":"refs/changes/32/951932/8","fetch":{"anonymous http":{"url":"https://review.opendev.org/openstack/openstack-ansible-lxc_hosts","ref":"refs/changes/32/951932/8","commands":{"Checkout":"git fetch https://review.opendev.org/openstack/openstack-ansible-lxc_hosts refs/changes/32/951932/8 \u0026\u0026 git checkout FETCH_HEAD","Cherry Pick":"git fetch https://review.opendev.org/openstack/openstack-ansible-lxc_hosts refs/changes/32/951932/8 \u0026\u0026 git cherry-pick FETCH_HEAD","Format Patch":"git fetch https://review.opendev.org/openstack/openstack-ansible-lxc_hosts refs/changes/32/951932/8 \u0026\u0026 git format-patch -1 --stdout FETCH_HEAD","Pull":"git pull https://review.opendev.org/openstack/openstack-ansible-lxc_hosts refs/changes/32/951932/8"}}},"commit":{"parents":[{"commit":"b69a166731f7d23849100c6ad9ff999ddf0ca9d3","subject":"Do not use distro templates by default","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/openstack-ansible-lxc_hosts/commit/b69a166731f7d23849100c6ad9ff999ddf0ca9d3"}]}],"author":{"name":"Dmitriy Rabotyagov","email":"dmitriy.rabotyagov@citynetwork.eu","date":"2022-05-27 18:19:35.000000000","tz":120},"committer":{"name":"Dmitriy Rabotyagov","email":"dmitriy.rabotyagov@cleura.com","date":"2025-06-06 18:59:31.000000000","tz":0},"subject":"Do not install COPR repo for CentOS LXC","message":"Do not install COPR repo for CentOS LXC\n\nEPEL does contain LXC package for CentOS 8 and 9 Stream, so we can\ninstall LXC from there. For CentOS 9 COPR still needed for lxc-templates\n\nWith that we also add control of what packages are going to be installed\nfrom EPEL.\n\nChange-Id: I60f1b04c2378c91905440e8de223c29dd6da4d17\n(cherry picked from commit c84b23b25851a8fca67a5a6e70d8bc687d5832d5)\n","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/openstack-ansible-lxc_hosts/commit/bca3f2b11a73ae1e82c93eeabf06f1b11179d475"}],"resolve_conflicts_web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/openstack-ansible-lxc_hosts/commit/bca3f2b11a73ae1e82c93eeabf06f1b11179d475"}]},"branch":"refs/heads/stable/2025.1"},"1de59fb340868a7fc6b734556a23c2d312dcc469":{"kind":"REWORK","_number":9,"created":"2025-06-06 20:20:51.000000000","uploader":{"_account_id":28619,"name":"Dmitriy Rabotyagov","email":"noonedeadpunk@gmail.com","username":"noonedeadpunk"},"ref":"refs/changes/32/951932/9","fetch":{"anonymous http":{"url":"https://review.opendev.org/openstack/openstack-ansible-lxc_hosts","ref":"refs/changes/32/951932/9","commands":{"Checkout":"git fetch https://review.opendev.org/openstack/openstack-ansible-lxc_hosts refs/changes/32/951932/9 \u0026\u0026 git checkout FETCH_HEAD","Cherry Pick":"git fetch https://review.opendev.org/openstack/openstack-ansible-lxc_hosts refs/changes/32/951932/9 \u0026\u0026 git cherry-pick FETCH_HEAD","Format Patch":"git fetch https://review.opendev.org/openstack/openstack-ansible-lxc_hosts refs/changes/32/951932/9 \u0026\u0026 git format-patch -1 --stdout FETCH_HEAD","Pull":"git pull https://review.opendev.org/openstack/openstack-ansible-lxc_hosts refs/changes/32/951932/9"}}},"commit":{"parents":[{"commit":"b69a166731f7d23849100c6ad9ff999ddf0ca9d3","subject":"Do not use distro templates by default","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/openstack-ansible-lxc_hosts/commit/b69a166731f7d23849100c6ad9ff999ddf0ca9d3"}]}],"author":{"name":"Dmitriy Rabotyagov","email":"dmitriy.rabotyagov@citynetwork.eu","date":"2022-05-27 18:19:35.000000000","tz":120},"committer":{"name":"Dmitriy Rabotyagov","email":"dmitriy.rabotyagov@cleura.com","date":"2025-06-06 20:20:51.000000000","tz":0},"subject":"Do not install COPR repo for CentOS LXC","message":"Do not install COPR repo for CentOS LXC\n\nEPEL does contain LXC package for CentOS 8 and 9 Stream, so we can\ninstall LXC from there. For CentOS 9 COPR still needed for lxc-templates\n\nWith that we also add control of what packages are going to be installed\nfrom EPEL.\n\nChange-Id: I60f1b04c2378c91905440e8de223c29dd6da4d17\n(cherry picked from commit c84b23b25851a8fca67a5a6e70d8bc687d5832d5)\n","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/openstack-ansible-lxc_hosts/commit/1de59fb340868a7fc6b734556a23c2d312dcc469"}],"resolve_conflicts_web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/openstack-ansible-lxc_hosts/commit/1de59fb340868a7fc6b734556a23c2d312dcc469"}]},"branch":"refs/heads/stable/2025.1"}},"requirements":[],"submit_records":[{"rule_name":"gerrit~DefaultSubmitRule","status":"CLOSED","labels":[{"label":"Verified","status":"MAY","applied_by":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]}},{"label":"Code-Review","status":"MAY","applied_by":{"_account_id":34411,"name":"Neil Hanlon","email":"neil@shrug.pw","username":"nhanlon"}},{"label":"Workflow","status":"MAY","applied_by":{"_account_id":34411,"name":"Neil Hanlon","email":"neil@shrug.pw","username":"nhanlon"}},{"label":"Backport-Candidate","status":"MAY"}]}],"submit_requirements":[{"name":"Verified","description":"Verified in gate by CI","status":"SATISFIED","is_legacy":false,"submittability_expression_result":{"expression":"label:Verified\u003dMAX AND -label:Verified\u003dMIN","fulfilled":true,"status":"PASS","passing_atoms":["label:Verified\u003dMAX"],"failing_atoms":["label:Verified\u003dMIN"],"atom_explanations":{}}},{"name":"Backport-Candidate","description":"Backport candidate status","status":"NOT_APPLICABLE","is_legacy":false,"applicability_expression_result":{"fulfilled":false,"status":"FAIL"},"submittability_expression_result":{"expression":"is:true","fulfilled":true,"status":"NOT_EVALUATED","passing_atoms":[],"failing_atoms":[],"atom_explanations":{}}},{"name":"Code-Review","description":"Code reviewed by core reviewer","status":"SATISFIED","is_legacy":false,"submittability_expression_result":{"expression":"label:Code-Review\u003dMAX AND -label:Code-Review\u003dMIN","fulfilled":true,"status":"PASS","passing_atoms":["label:Code-Review\u003dMAX"],"failing_atoms":["label:Code-Review\u003dMIN"],"atom_explanations":{}}},{"name":"Workflow","description":"Approved for gate by core reviewer","status":"SATISFIED","is_legacy":false,"submittability_expression_result":{"expression":"label:Workflow\u003dMAX AND -label:Workflow\u003dMIN","fulfilled":true,"status":"PASS","passing_atoms":["label:Workflow\u003dMAX"],"failing_atoms":["label:Workflow\u003dMIN"],"atom_explanations":{}}}]}
