)]}'
{"id":"openstack%2Fopenstack-ansible-lxc_hosts~972848","triplet_id":"openstack%2Fopenstack-ansible-lxc_hosts~master~Idccfd72ce57e5fdd0f7ad6e2def5dd153f8ddbbe","project":"openstack/openstack-ansible-lxc_hosts","branch":"master","attention_set":{},"removed_from_attention_set":{"31542":{"account":{"_account_id":31542,"name":"Andrew Bonney","email":"andrew.bonney@bbc.co.uk","username":"andrewbonney"},"last_update":"2026-01-12 08:19:25.000000000","reason":"\u003cGERRIT_ACCOUNT_31542\u003e replied on the change","reason_account":{"_account_id":31542,"name":"Andrew Bonney","email":"andrew.bonney@bbc.co.uk","username":"andrewbonney"}},"28619":{"account":{"_account_id":28619,"name":"Dmitriy Rabotyagov","email":"noonedeadpunk@gmail.com","username":"noonedeadpunk"},"last_update":"2026-01-23 17:25:52.000000000","reason":"Change was submitted"},"25023":{"account":{"_account_id":25023,"name":"Jonathan Rosser","email":"jonathan.rosser@rd.bbc.co.uk","username":"jrosser"},"last_update":"2026-01-23 17:25:52.000000000","reason":"Change was submitted"}},"hashtags":[],"change_id":"Idccfd72ce57e5fdd0f7ad6e2def5dd153f8ddbbe","subject":"Revert \"Temporary disable RHEL jobs\"","status":"MERGED","created":"2026-01-09 11:44:15.000000000","updated":"2026-01-23 17:27:50.000000000","submitted":"2026-01-23 17:25:52.000000000","submitter":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"total_comment_count":8,"unresolved_comment_count":0,"has_review_started":true,"submission_id":"972848","meta_rev_id":"9cdaa2287f4c589770946cb28dfddebc73ee8a0d","_number":972848,"virtual_id_number":972848,"owner":{"_account_id":28619,"name":"Dmitriy Rabotyagov","email":"noonedeadpunk@gmail.com","username":"noonedeadpunk"},"actions":{},"labels":{"Verified":{"approved":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"all":[{"value":0,"_account_id":37632,"name":"Dmitriy Chubinidze","email":"dcu995@gmail.com","username":"chubinidzedr"},{"value":0,"_account_id":31542,"name":"Andrew Bonney","email":"andrew.bonney@bbc.co.uk","username":"andrewbonney"},{"value":0,"_account_id":28619,"name":"Dmitriy Rabotyagov","email":"noonedeadpunk@gmail.com","username":"noonedeadpunk"},{"tag":"autogenerated:zuul:gate","value":2,"date":"2026-01-23 17:25:52.000000000","permitted_voting_range":{"min":2,"max":2},"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},{"value":0,"_account_id":37598,"name":"Ivan Anfimov","display_name":"Ivan Anfimov","email":"lazekteam@gmail.com","username":"anfimovir"},{"value":0,"_account_id":25023,"name":"Jonathan Rosser","email":"jonathan.rosser@rd.bbc.co.uk","username":"jrosser"}],"values":{"-2":"Fails","-1":"Doesn\u0027t seem to work"," 0":"No score","+1":"Works for me","+2":"Verified"},"description":"","default_value":0,"optional":true},"Code-Review":{"approved":{"_account_id":31542,"name":"Andrew Bonney","email":"andrew.bonney@bbc.co.uk","username":"andrewbonney"},"recommended":{"_account_id":37632,"name":"Dmitriy Chubinidze","email":"dcu995@gmail.com","username":"chubinidzedr"},"all":[{"value":1,"date":"2026-01-09 20:59:35.000000000","permitted_voting_range":{"min":1,"max":1},"_account_id":37632,"name":"Dmitriy Chubinidze","email":"dcu995@gmail.com","username":"chubinidzedr"},{"value":2,"date":"2026-01-12 08:19:25.000000000","permitted_voting_range":{"min":2,"max":2},"_account_id":31542,"name":"Andrew Bonney","email":"andrew.bonney@bbc.co.uk","username":"andrewbonney"},{"value":0,"_account_id":28619,"name":"Dmitriy Rabotyagov","email":"noonedeadpunk@gmail.com","username":"noonedeadpunk"},{"value":0,"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},{"value":1,"date":"2026-01-09 13:44:12.000000000","permitted_voting_range":{"min":1,"max":2},"_account_id":37598,"name":"Ivan Anfimov","display_name":"Ivan Anfimov","email":"lazekteam@gmail.com","username":"anfimovir"},{"value":2,"date":"2026-01-12 10:00:23.000000000","permitted_voting_range":{"min":2,"max":2},"_account_id":25023,"name":"Jonathan Rosser","email":"jonathan.rosser@rd.bbc.co.uk","username":"jrosser"}],"values":{"-2":"Do not merge","-1":"This patch needs further work before it can be merged"," 0":"No score","+1":"Looks good to me, but someone else must approve","+2":"Looks good to me (core reviewer)"},"description":"","value":1,"default_value":0,"optional":true},"Workflow":{"approved":{"_account_id":28619,"name":"Dmitriy Rabotyagov","email":"noonedeadpunk@gmail.com","username":"noonedeadpunk"},"all":[{"value":0,"_account_id":37632,"name":"Dmitriy Chubinidze","email":"dcu995@gmail.com","username":"chubinidzedr"},{"value":0,"_account_id":31542,"name":"Andrew Bonney","email":"andrew.bonney@bbc.co.uk","username":"andrewbonney"},{"value":1,"date":"2026-01-22 14:38:56.000000000","permitted_voting_range":{"min":1,"max":1},"_account_id":28619,"name":"Dmitriy Rabotyagov","email":"noonedeadpunk@gmail.com","username":"noonedeadpunk"},{"value":0,"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},{"value":0,"_account_id":37598,"name":"Ivan Anfimov","display_name":"Ivan Anfimov","email":"lazekteam@gmail.com","username":"anfimovir"},{"value":0,"date":"2026-01-12 10:39:17.000000000","permitted_voting_range":{"min":0,"max":1},"_account_id":25023,"name":"Jonathan Rosser","email":"jonathan.rosser@rd.bbc.co.uk","username":"jrosser"}],"values":{"-1":"Work in progress"," 0":"Ready for reviews","+1":"Approved"},"description":"","default_value":0,"optional":true},"Backport-Candidate":{"all":[{"value":0,"_account_id":37632,"name":"Dmitriy Chubinidze","email":"dcu995@gmail.com","username":"chubinidzedr"},{"value":0,"_account_id":31542,"name":"Andrew Bonney","email":"andrew.bonney@bbc.co.uk","username":"andrewbonney"},{"value":0,"_account_id":28619,"name":"Dmitriy Rabotyagov","email":"noonedeadpunk@gmail.com","username":"noonedeadpunk"},{"value":0,"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},{"value":0,"_account_id":37598,"name":"Ivan Anfimov","display_name":"Ivan Anfimov","email":"lazekteam@gmail.com","username":"anfimovir"},{"value":0,"_account_id":25023,"name":"Jonathan Rosser","email":"jonathan.rosser@rd.bbc.co.uk","username":"jrosser"}],"values":{"-1":"Not A Backport Candidate"," 0":"Backport Review Needed","+1":"Proposed Backport"},"description":"","default_value":0,"optional":true}},"removable_reviewers":[],"reviewers":{"REVIEWER":[{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},{"_account_id":25023,"name":"Jonathan Rosser","email":"jonathan.rosser@rd.bbc.co.uk","username":"jrosser"},{"_account_id":28619,"name":"Dmitriy Rabotyagov","email":"noonedeadpunk@gmail.com","username":"noonedeadpunk"},{"_account_id":31542,"name":"Andrew Bonney","email":"andrew.bonney@bbc.co.uk","username":"andrewbonney"},{"_account_id":37598,"name":"Ivan Anfimov","display_name":"Ivan Anfimov","email":"lazekteam@gmail.com","username":"anfimovir"},{"_account_id":37632,"name":"Dmitriy Chubinidze","email":"dcu995@gmail.com","username":"chubinidzedr"}]},"pending_reviewers":{},"reviewer_updates":[{"updated":"2026-01-09 13:09:47.000000000","updated_by":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"reviewer":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"state":"REVIEWER"},{"updated":"2026-01-09 13:44:12.000000000","updated_by":{"_account_id":37598,"name":"Ivan Anfimov","display_name":"Ivan Anfimov","email":"lazekteam@gmail.com","username":"anfimovir"},"reviewer":{"_account_id":37598,"name":"Ivan Anfimov","display_name":"Ivan Anfimov","email":"lazekteam@gmail.com","username":"anfimovir"},"state":"REVIEWER"},{"updated":"2026-01-09 13:44:12.000000000","updated_by":{"_account_id":37598,"name":"Ivan Anfimov","display_name":"Ivan Anfimov","email":"lazekteam@gmail.com","username":"anfimovir"},"reviewer":{"_account_id":31542,"name":"Andrew Bonney","email":"andrew.bonney@bbc.co.uk","username":"andrewbonney"},"state":"REVIEWER"},{"updated":"2026-01-09 13:44:12.000000000","updated_by":{"_account_id":37598,"name":"Ivan Anfimov","display_name":"Ivan Anfimov","email":"lazekteam@gmail.com","username":"anfimovir"},"reviewer":{"_account_id":25023,"name":"Jonathan Rosser","email":"jonathan.rosser@rd.bbc.co.uk","username":"jrosser"},"state":"REVIEWER"},{"updated":"2026-01-09 20:59:35.000000000","updated_by":{"_account_id":37632,"name":"Dmitriy Chubinidze","email":"dcu995@gmail.com","username":"chubinidzedr"},"reviewer":{"_account_id":37632,"name":"Dmitriy Chubinidze","email":"dcu995@gmail.com","username":"chubinidzedr"},"state":"REVIEWER"}],"messages":[{"id":"0d3264deb77867c602325ddbbc3cd2f4fb85c81f","tag":"autogenerated:gerrit:newPatchSet","author":{"_account_id":28619,"name":"Dmitriy Rabotyagov","email":"noonedeadpunk@gmail.com","username":"noonedeadpunk"},"date":"2026-01-09 11:44:15.000000000","message":"Uploaded patch set 1.","accounts_in_message":[],"_revision_number":1},{"id":"69e873ab3c031cebd1055ab4ef4805b4d06aea34","tag":"autogenerated:zuul:check","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2026-01-09 13:09:47.000000000","message":"Patch Set 1: Verified+1\n\nBuild succeeded (check pipeline).\nhttps://zuul.opendev.org/t/openstack/buildset/99c5ca69b78f49668710b94553da8bd4\n\n- openstack-ansible-linters-ubuntu-noble https://zuul.opendev.org/t/openstack/build/2ff7508be2a74dada7638c04db77d33e : SUCCESS in 9m 03s\n- openstack-tox-docs https://zuul.opendev.org/t/openstack/build/1994c82c08f34fb6bd7f5e94f75f39d8 : SUCCESS in 5m 10s\n- openstack-ansible-deploy-infra_lxc-centos-10-stream https://zuul.opendev.org/t/openstack/build/34574a5aa4354dab9b8dddd662006b43 : SUCCESS in 1h 01m 40s\n- openstack-ansible-deploy-infra_lxc-debian-bookworm https://zuul.opendev.org/t/openstack/build/80b6f6d693104793989380fb633dbf31 : SUCCESS in 1h 03m 13s\n- openstack-ansible-deploy-infra_lxc-debian-trixie https://zuul.opendev.org/t/openstack/build/769054e0a9384401bf76caa39fb9002d : SUCCESS in 48m 00s\n- openstack-ansible-deploy-infra_lxc-rockylinux-10 https://zuul.opendev.org/t/openstack/build/c6b8e408a59543fc92a7a00a1d2c0325 : SUCCESS in 58m 58s\n- openstack-ansible-deploy-infra_lxc_tls-rockylinux-10 https://zuul.opendev.org/t/openstack/build/290ca430fed6404983e32bb5a01b93ce : SUCCESS in 50m 41s\n- openstack-ansible-deploy-infra_lxc-ubuntu-noble https://zuul.opendev.org/t/openstack/build/9eed08c91b384daa81c370b1a2310e9a : SUCCESS in 1h 20m 59s\n- openstack-ansible-deploy-infra_lxc_stepca-ubuntu-noble https://zuul.opendev.org/t/openstack/build/d9c6644af1174bd29394682d7da7b7af : SUCCESS in 49m 01s\n- openstack-ansible-deploy-infra_lxc_tls-ubuntu-noble https://zuul.opendev.org/t/openstack/build/fc2fd9dae5a54f05bf949292a9b911a2 : SUCCESS in 45m 42s\n- openstack-ansible-upgrade-infra_lxc-ubuntu-noble https://zuul.opendev.org/t/openstack/build/72703ec202804f299496714fe237d9a7 : SUCCESS in 55m 48s\n- openstack-ansible-deploy-hosts_lxc_zfs-ubuntu-noble https://zuul.opendev.org/t/openstack/build/b02155b2008742cbb1125058198c52ee : SUCCESS in 16m 50s\n- openstack-ansible-deploy-hosts_lxc_btrfs-ubuntu-noble https://zuul.opendev.org/t/openstack/build/e7fc202511604bb3a7c8a337e81b5ef9 : SUCCESS in 26m 14s\n- openstack-ansible-deploy-hosts_lxc_overlayfs-ubuntu-noble https://zuul.opendev.org/t/openstack/build/d64e76f14df94ba2b5e30d12cf93de42 : SUCCESS in 26m 48s\n- openstack-ansible-deploy-hosts_lxc_btrfs-debian-bookworm https://zuul.opendev.org/t/openstack/build/3e1a6a7a92684e71a04723db5122b2e6 : SUCCESS in 17m 20s\n- openstack-ansible-deploy-hosts_lxc_overlayfs-debian-bookworm https://zuul.opendev.org/t/openstack/build/d616661279b5412aaadb9c32ffe8abaf : SUCCESS in 26m 28s\n- openstack-ansible-deploy-hosts_lxc_overlayfs-rockylinux-10 https://zuul.opendev.org/t/openstack/build/c5da75b5a57a431a8cb467a8571759c9 : SUCCESS in 17m 11s\n- openstack-ansible-deploy-hosts_lxc_btrfs-debian-trixie https://zuul.opendev.org/t/openstack/build/8f7ae3a2196949b5aa92b570101593e7 : SUCCESS in 17m 18s\n- openstack-ansible-deploy-hosts_lxc_overlayfs-debian-trixie https://zuul.opendev.org/t/openstack/build/657071887aed4ac0a8a6b951ceaff8e7 : SUCCESS in 26m 20s","accounts_in_message":[],"_revision_number":1},{"id":"37eb94f3e36093826403fc35d3a2800a42e4d390","author":{"_account_id":37598,"name":"Ivan Anfimov","display_name":"Ivan Anfimov","email":"lazekteam@gmail.com","username":"anfimovir"},"date":"2026-01-09 13:44:12.000000000","message":"Patch Set 1: Code-Review+1","accounts_in_message":[],"_revision_number":1},{"id":"cd1be20eb7a30998c515598d8595ca5c8bea7649","author":{"_account_id":37632,"name":"Dmitriy Chubinidze","email":"dcu995@gmail.com","username":"chubinidzedr"},"date":"2026-01-09 20:59:35.000000000","message":"Patch Set 1: Code-Review+1","accounts_in_message":[],"_revision_number":1},{"id":"7c5417795705c1a34db96c25673fc714cafded3b","author":{"_account_id":31542,"name":"Andrew Bonney","email":"andrew.bonney@bbc.co.uk","username":"andrewbonney"},"date":"2026-01-12 08:19:25.000000000","message":"Patch Set 1: Code-Review+2","accounts_in_message":[],"_revision_number":1},{"id":"45c1695217e38befa83dd6e55860e746326c26f2","author":{"_account_id":25023,"name":"Jonathan Rosser","email":"jonathan.rosser@rd.bbc.co.uk","username":"jrosser"},"date":"2026-01-12 10:00:23.000000000","message":"Patch Set 1: Code-Review+2 Workflow+1","accounts_in_message":[],"_revision_number":1},{"id":"17f5786d561c86578c9eeabc110cab477aeb53b7","tag":"autogenerated:zuul:gate","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2026-01-12 10:00:38.000000000","message":"Patch Set 1: Verified-2\n\nBuild failed (gate pipeline).  For information on how to proceed, see\nhttps://docs.opendev.org/opendev/infra-manual/latest/developers.html#automated-testing\n\nhttps://zuul.opendev.org/t/openstack/buildset/ee6ea5e746204a93b43914884644b369\n\n\nWarning:\n  Change 972497,2 in project openstack/openstack-ansible-rabbitmq_server does not share a change queue with 972848 in project openstack/openstack-ansible-lxc_hosts","accounts_in_message":[],"_revision_number":1},{"id":"e80c9b06aa20649d4332ef562cf4fdfe3eb47721","author":{"_account_id":25023,"name":"Jonathan Rosser","email":"jonathan.rosser@rd.bbc.co.uk","username":"jrosser"},"date":"2026-01-12 10:39:17.000000000","message":"Patch Set 1: -Workflow\n\n(1 comment)","accounts_in_message":[],"_revision_number":1},{"id":"36d1771272ff28dda01201372510f5d5a2c3ee64","tag":"autogenerated:zuul:check","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2026-01-12 12:07:10.000000000","message":"Patch Set 1: Verified-1\n\nBuild failed (check pipeline).  For information on how to proceed, see\nhttps://docs.opendev.org/opendev/infra-manual/latest/developers.html#automated-testing\nand https://docs.openstack.org/project-team-guide/testing.html#how-to-handle-test-failures\n\nhttps://zuul.opendev.org/t/openstack/buildset/7fd2b0e77382450a9571a152fbaf9d46\n\n- openstack-ansible-linters-ubuntu-noble https://zuul.opendev.org/t/openstack/build/28e6d3e0ed2a4ffdabe126868a3614ae : SUCCESS in 10m 59s\n- openstack-tox-docs https://zuul.opendev.org/t/openstack/build/f76ff988026747d58728050d9af9cd14 : SUCCESS in 4m 56s\n- openstack-ansible-deploy-infra_lxc-centos-10-stream https://zuul.opendev.org/t/openstack/build/5c55916a31a84adcbe75e9c675b81526 : SUCCESS in 1h 20m 47s\n- openstack-ansible-deploy-infra_lxc-debian-bookworm https://zuul.opendev.org/t/openstack/build/4bfa323ce78c4aad96a1854f72a5de46 : FAILURE in 1h 08m 38s\n- openstack-ansible-deploy-infra_lxc-debian-trixie https://zuul.opendev.org/t/openstack/build/775a96fe18bb485da27892fd82e54318 : SUCCESS in 1h 04m 49s\n- openstack-ansible-deploy-infra_lxc-rockylinux-10 https://zuul.opendev.org/t/openstack/build/a9e10762b4b84a1a8530b1b3ef236e6c : SUCCESS in 45m 08s\n- openstack-ansible-deploy-infra_lxc_tls-rockylinux-10 https://zuul.opendev.org/t/openstack/build/2466588cf3cf4746ac20d57eb20f1b85 : SUCCESS in 46m 59s\n- openstack-ansible-deploy-infra_lxc-ubuntu-noble https://zuul.opendev.org/t/openstack/build/11eab6453ebb40b1a26a162ca98ea80d : SUCCESS in 1h 11m 50s\n- openstack-ansible-deploy-infra_lxc_stepca-ubuntu-noble https://zuul.opendev.org/t/openstack/build/2219e912b24b464ab261d113454bd086 : SUCCESS in 1h 14m 51s\n- openstack-ansible-deploy-infra_lxc_tls-ubuntu-noble https://zuul.opendev.org/t/openstack/build/19c08e69a4f642c1a41aca67fd941bb2 : SUCCESS in 1h 15m 40s\n- openstack-ansible-upgrade-infra_lxc-ubuntu-noble https://zuul.opendev.org/t/openstack/build/0157420634564a02b0dd5eb6104c700d : SUCCESS in 1h 21m 05s\n- openstack-ansible-deploy-hosts_lxc_zfs-ubuntu-noble https://zuul.opendev.org/t/openstack/build/6fb2eb2e033243ba844e4be4350e4982 : SUCCESS in 14m 08s\n- openstack-ansible-deploy-hosts_lxc_btrfs-ubuntu-noble https://zuul.opendev.org/t/openstack/build/e2b3db4a495048d08efd67084a48bf60 : SUCCESS in 21m 47s\n- openstack-ansible-deploy-hosts_lxc_overlayfs-ubuntu-noble https://zuul.opendev.org/t/openstack/build/4ca77f933d4a42038e081f791cb28d20 : SUCCESS in 22m 59s\n- openstack-ansible-deploy-hosts_lxc_btrfs-debian-bookworm https://zuul.opendev.org/t/openstack/build/11b258555fdf42529c5de399e93b9572 : SUCCESS in 22m 59s\n- openstack-ansible-deploy-hosts_lxc_overlayfs-debian-bookworm https://zuul.opendev.org/t/openstack/build/db0841f3c507402e810a4e58499f0679 : SUCCESS in 20m 58s\n- openstack-ansible-deploy-hosts_lxc_overlayfs-rockylinux-10 https://zuul.opendev.org/t/openstack/build/c139914a82de4eb18e9349b427726708 : SUCCESS in 27m 54s\n- openstack-ansible-deploy-hosts_lxc_btrfs-debian-trixie https://zuul.opendev.org/t/openstack/build/e7dd922d5b5042188a7078d66602b2cb : SUCCESS in 16m 58s\n- openstack-ansible-deploy-hosts_lxc_overlayfs-debian-trixie https://zuul.opendev.org/t/openstack/build/9d933a4446f54d3ba188596a20c74abc : SUCCESS in 21m 19s","accounts_in_message":[],"_revision_number":1},{"id":"b0a7f0a884cba94cd1804e7f91f4d23bd29aea62","author":{"_account_id":37598,"name":"Ivan Anfimov","display_name":"Ivan Anfimov","email":"lazekteam@gmail.com","username":"anfimovir"},"date":"2026-01-19 07:28:12.000000000","message":"Patch Set 1:\n\n(1 comment)","accounts_in_message":[],"_revision_number":1},{"id":"694887cba8b4afef16b08f1b1ab225b5a9471760","author":{"_account_id":37632,"name":"Dmitriy Chubinidze","email":"dcu995@gmail.com","username":"chubinidzedr"},"date":"2026-01-19 07:28:55.000000000","message":"Patch Set 1:\n\n(1 comment)","accounts_in_message":[],"_revision_number":1},{"id":"4d091ba87de5cbb193e049df1a3472e74248b40e","tag":"autogenerated:zuul:check","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2026-01-19 09:26:09.000000000","message":"Patch Set 1: Verified+1\n\nBuild succeeded (check pipeline).\nhttps://zuul.opendev.org/t/openstack/buildset/5b11da8aaa074fa79867c1e2458818f1\n\n- openstack-ansible-linters-ubuntu-noble https://zuul.opendev.org/t/openstack/build/a5207a48337542c1a722ad6389cee159 : SUCCESS in 10m 02s\n- openstack-tox-docs https://zuul.opendev.org/t/openstack/build/663a4a9ed4494e248f2b3c2d613d711c : SUCCESS in 4m 56s\n- openstack-ansible-deploy-infra_lxc-centos-10-stream https://zuul.opendev.org/t/openstack/build/ddf14819870748a28917a41fcbde1de2 : SUCCESS in 1h 20m 49s\n- openstack-ansible-deploy-infra_lxc-debian-bookworm https://zuul.opendev.org/t/openstack/build/c3fe02ada3294eda8663882684d7c2fe : SUCCESS in 44m 37s\n- openstack-ansible-deploy-infra_lxc-debian-trixie https://zuul.opendev.org/t/openstack/build/c2f2e950f1014c83aeb531b47421b020 : SUCCESS in 48m 59s\n- openstack-ansible-deploy-infra_lxc-rockylinux-10 https://zuul.opendev.org/t/openstack/build/c146ac1f45c34f6eabb373277a5b7d40 : SUCCESS in 48m 24s\n- openstack-ansible-deploy-infra_lxc_tls-rockylinux-10 https://zuul.opendev.org/t/openstack/build/1724138b7172419b8cc318538bb5ee14 : SUCCESS in 1h 09m 42s\n- openstack-ansible-deploy-infra_lxc-ubuntu-noble https://zuul.opendev.org/t/openstack/build/f4c2301c9fbb45d797fd2057af58182e : SUCCESS in 1h 25m 05s\n- openstack-ansible-deploy-infra_lxc_stepca-ubuntu-noble https://zuul.opendev.org/t/openstack/build/29aa8bc11b494362991a4f0704fcd4c7 : SUCCESS in 1h 15m 23s\n- openstack-ansible-deploy-infra_lxc_tls-ubuntu-noble https://zuul.opendev.org/t/openstack/build/f06fdc4c2f644ed49110f22121003b09 : SUCCESS in 47m 48s\n- openstack-ansible-upgrade-infra_lxc-ubuntu-noble https://zuul.opendev.org/t/openstack/build/ce661b09f821478b964ebb01fbb14fce : SUCCESS in 1h 45m 58s\n- openstack-ansible-deploy-hosts_lxc_zfs-ubuntu-noble https://zuul.opendev.org/t/openstack/build/da23a58f26354dba94deb92e44efc4bb : SUCCESS in 23m 40s\n- openstack-ansible-deploy-hosts_lxc_btrfs-ubuntu-noble https://zuul.opendev.org/t/openstack/build/c55688b0d22748e98e715b192fb0ba15 : SUCCESS in 15m 35s\n- openstack-ansible-deploy-hosts_lxc_overlayfs-ubuntu-noble https://zuul.opendev.org/t/openstack/build/a489503163954ec5b613c7cd987e0faf : SUCCESS in 22m 44s\n- openstack-ansible-deploy-hosts_lxc_btrfs-debian-bookworm https://zuul.opendev.org/t/openstack/build/871e1c032eed4152b3379d639b8f7662 : SUCCESS in 23m 39s\n- openstack-ansible-deploy-hosts_lxc_overlayfs-debian-bookworm https://zuul.opendev.org/t/openstack/build/2bc1883c42d44e5c8c407cde97612d25 : SUCCESS in 22m 34s\n- openstack-ansible-deploy-hosts_lxc_overlayfs-rockylinux-10 https://zuul.opendev.org/t/openstack/build/1b57bd45b4f14a7788478a771794004f : SUCCESS in 19m 03s\n- openstack-ansible-deploy-hosts_lxc_btrfs-debian-trixie https://zuul.opendev.org/t/openstack/build/dc9fc188fb32476e80135418dbd2a9ad : SUCCESS in 22m 52s\n- openstack-ansible-deploy-hosts_lxc_overlayfs-debian-trixie https://zuul.opendev.org/t/openstack/build/1c084b14a85c4a0ca4981073692c5999 : SUCCESS in 20m 29s","accounts_in_message":[],"_revision_number":1},{"id":"830ffb8f744ac29882daa8234d784a99b883c9c3","author":{"_account_id":37598,"name":"Ivan Anfimov","display_name":"Ivan Anfimov","email":"lazekteam@gmail.com","username":"anfimovir"},"date":"2026-01-19 09:28:14.000000000","message":"Patch Set 1:\n\n(1 comment)","accounts_in_message":[],"_revision_number":1},{"id":"292169155f64c1becab1bdfb4b63486a3558c803","author":{"_account_id":37598,"name":"Ivan Anfimov","display_name":"Ivan Anfimov","email":"lazekteam@gmail.com","username":"anfimovir"},"date":"2026-01-22 12:33:05.000000000","message":"Patch Set 1:\n\n(1 comment)","accounts_in_message":[],"_revision_number":1},{"id":"e61af4415d2c7b75dd2c12d28111179c83a3f149","author":{"_account_id":28619,"name":"Dmitriy Rabotyagov","email":"noonedeadpunk@gmail.com","username":"noonedeadpunk"},"date":"2026-01-22 14:38:56.000000000","message":"Patch Set 1: Workflow+1","accounts_in_message":[],"_revision_number":1},{"id":"2c9302af442176a3ef33c7da0033124cecd7b797","tag":"autogenerated:zuul:gate","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2026-01-22 14:41:32.000000000","message":"Patch Set 1: -Verified\n\nStarting gate jobs.","accounts_in_message":[],"_revision_number":1},{"id":"c57857c956eb719d4bb68c254515bf3d887539a1","tag":"autogenerated:zuul:gate","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2026-01-22 17:15:15.000000000","message":"Patch Set 1: Verified-2\n\nBuild failed (gate pipeline).  For information on how to proceed, see\nhttps://docs.opendev.org/opendev/infra-manual/latest/developers.html#automated-testing\n\nhttps://zuul.opendev.org/t/openstack/buildset/488324e0bbb74295a3b920f4038de88c\n\n- openstack-ansible-linters-ubuntu-noble https://zuul.opendev.org/t/openstack/build/879eb62d34984d53ae4af9eaff226b23 : SUCCESS in 9m 22s\n- openstack-tox-docs https://zuul.opendev.org/t/openstack/build/963cac6475134f5e8e073af3d2dfe3c8 : SUCCESS in 6m 34s\n- openstack-ansible-deploy-infra_lxc-centos-10-stream https://zuul.opendev.org/t/openstack/build/a0e907920c494e93b92d8001b382b9de : FAILURE in 28m 53s\n- openstack-ansible-deploy-infra_lxc-debian-bookworm https://zuul.opendev.org/t/openstack/build/03f4fe9fad794824afbd16bdccfaee67 : SUCCESS in 1h 16m 27s\n- openstack-ansible-deploy-infra_lxc-debian-trixie https://zuul.opendev.org/t/openstack/build/cf3349aa17ca46efbc641cfb496e5e1c : SUCCESS in 1h 23m 55s\n- openstack-ansible-deploy-infra_lxc-rockylinux-10 https://zuul.opendev.org/t/openstack/build/34377f79035641d2ab3c1b7b22188943 : FAILURE in 31m 28s\n- openstack-ansible-deploy-infra_lxc_tls-rockylinux-10 https://zuul.opendev.org/t/openstack/build/ad13a114503646bbaff472906705b29e : FAILURE in 38m 40s\n- openstack-ansible-deploy-infra_lxc-ubuntu-noble https://zuul.opendev.org/t/openstack/build/964c8fa534774798b34162b87dbdcd81 : SUCCESS in 1h 17m 17s\n- openstack-ansible-deploy-infra_lxc_stepca-ubuntu-noble https://zuul.opendev.org/t/openstack/build/eaa70d9a44ca420fbc238e7aa9f6540d : SUCCESS in 44m 41s\n- openstack-ansible-deploy-infra_lxc_tls-ubuntu-noble https://zuul.opendev.org/t/openstack/build/6a5f5eb7d0144cafa7bcb3e507311f45 : SUCCESS in 1h 24m 30s\n- openstack-ansible-upgrade-infra_lxc-ubuntu-noble https://zuul.opendev.org/t/openstack/build/cf42f3483cde43978b1fc1855e526a4e : SUCCESS in 1h 37m 11s\n- openstack-ansible-deploy-hosts_lxc_zfs-ubuntu-noble https://zuul.opendev.org/t/openstack/build/93394ec430b54131bfd9426340241702 : SUCCESS in 28m 22s\n- openstack-ansible-deploy-hosts_lxc_btrfs-ubuntu-noble https://zuul.opendev.org/t/openstack/build/97e55278d73648ada9caec45cb3d0c5e : SUCCESS in 27m 56s\n- openstack-ansible-deploy-hosts_lxc_overlayfs-ubuntu-noble https://zuul.opendev.org/t/openstack/build/24e33e3e5d0a4c42b47a215900068e25 : SUCCESS in 19m 39s\n- openstack-ansible-deploy-hosts_lxc_btrfs-debian-bookworm https://zuul.opendev.org/t/openstack/build/d9032cd23dda4cc1a11934cac708f983 : SUCCESS in 27m 19s\n- openstack-ansible-deploy-hosts_lxc_overlayfs-debian-bookworm https://zuul.opendev.org/t/openstack/build/284b5f18373b411992a1a679975070a7 : SUCCESS in 22m 14s\n- openstack-ansible-deploy-hosts_lxc_btrfs-debian-trixie https://zuul.opendev.org/t/openstack/build/597e95e1a2874466bd04fc8707a2d2e3 : SUCCESS in 20m 39s\n- openstack-ansible-deploy-hosts_lxc_overlayfs-debian-trixie https://zuul.opendev.org/t/openstack/build/d8c34b51246046f1ad45d1e9293e4c9d : SUCCESS in 18m 26s\n- openstack-ansible-deploy-hosts_lxc_overlayfs-rockylinux-10 https://zuul.opendev.org/t/openstack/build/89e68a4dd6784b42906ae9f420b0978a : SUCCESS in 33m 30s","accounts_in_message":[],"_revision_number":1},{"id":"582921982db69409c558bd117d0e7779acd89b09","author":{"_account_id":37598,"name":"Ivan Anfimov","display_name":"Ivan Anfimov","email":"lazekteam@gmail.com","username":"anfimovir"},"date":"2026-01-22 21:24:43.000000000","message":"Patch Set 1:\n\n(1 comment)","accounts_in_message":[],"_revision_number":1},{"id":"323557cc1cdd13a29ea71aea6cb1177fa9eb6028","tag":"autogenerated:zuul:check","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2026-01-22 23:25:44.000000000","message":"Patch Set 1: Verified+1\n\nBuild succeeded (check pipeline).\nhttps://zuul.opendev.org/t/openstack/buildset/df3cb2dfc10a46e3b1cd672755655275\n\n- openstack-ansible-linters-ubuntu-noble https://zuul.opendev.org/t/openstack/build/e9b63a845a5345dea017b43f79da8283 : SUCCESS in 8m 11s\n- openstack-tox-docs https://zuul.opendev.org/t/openstack/build/9b6b6ec2e07f4411940c4a0383d41c80 : SUCCESS in 4m 47s\n- openstack-ansible-deploy-infra_lxc-centos-10-stream https://zuul.opendev.org/t/openstack/build/d26f5f417ef142d6b82f7845c8dd0fad : SUCCESS in 1h 07m 50s\n- openstack-ansible-deploy-infra_lxc-debian-bookworm https://zuul.opendev.org/t/openstack/build/8141adf948d644479375947e7bc2fd94 : SUCCESS in 1h 15m 20s\n- openstack-ansible-deploy-infra_lxc-debian-trixie https://zuul.opendev.org/t/openstack/build/d80ce5167eff4506a5dbc26fee93f778 : SUCCESS in 1h 15m 10s\n- openstack-ansible-deploy-infra_lxc-rockylinux-10 https://zuul.opendev.org/t/openstack/build/1d31e0fe2e6c4d9d82525b508c514f77 : SUCCESS in 50m 11s\n- openstack-ansible-deploy-infra_lxc_tls-rockylinux-10 https://zuul.opendev.org/t/openstack/build/7faefbc2a999405482851916a3a95b4a : SUCCESS in 50m 07s\n- openstack-ansible-deploy-infra_lxc-ubuntu-noble https://zuul.opendev.org/t/openstack/build/8be87322a16e43238d70bb7e41ba115a : SUCCESS in 45m 46s\n- openstack-ansible-deploy-infra_lxc_stepca-ubuntu-noble https://zuul.opendev.org/t/openstack/build/761dcc8443e8410c9f854004eee739f4 : SUCCESS in 1h 16m 24s\n- openstack-ansible-deploy-infra_lxc_tls-ubuntu-noble https://zuul.opendev.org/t/openstack/build/e1b8b0c59137420098da3f619c3e893a : SUCCESS in 43m 49s\n- openstack-ansible-upgrade-infra_lxc-ubuntu-noble https://zuul.opendev.org/t/openstack/build/d4734251c4ec474ba4d57f81a8981e70 : SUCCESS in 1h 52m 25s\n- openstack-ansible-deploy-hosts_lxc_zfs-ubuntu-noble https://zuul.opendev.org/t/openstack/build/b7e43ee5fbb14ed589a6b9b1092eaeff : SUCCESS in 24m 31s\n- openstack-ansible-deploy-hosts_lxc_btrfs-ubuntu-noble https://zuul.opendev.org/t/openstack/build/f93f6ea3cded49e2a0f27ff8d3952220 : SUCCESS in 17m 07s\n- openstack-ansible-deploy-hosts_lxc_overlayfs-ubuntu-noble https://zuul.opendev.org/t/openstack/build/8fab5b257f9f47069abf86a52a381c55 : SUCCESS in 25m 06s\n- openstack-ansible-deploy-hosts_lxc_btrfs-debian-bookworm https://zuul.opendev.org/t/openstack/build/02409f0bc3b346eeb34bfc67a4be6683 : SUCCESS in 23m 00s\n- openstack-ansible-deploy-hosts_lxc_overlayfs-debian-bookworm https://zuul.opendev.org/t/openstack/build/c774c368e91d462d922dbc4bf5034d7b : SUCCESS in 24m 17s\n- openstack-ansible-deploy-hosts_lxc_overlayfs-rockylinux-10 https://zuul.opendev.org/t/openstack/build/7e84d33a54424c88b22c69e3a570df25 : SUCCESS in 21m 16s\n- openstack-ansible-deploy-hosts_lxc_btrfs-debian-trixie https://zuul.opendev.org/t/openstack/build/04aba8f674df4a2e88ac1bb3808270cc : SUCCESS in 21m 14s\n- openstack-ansible-deploy-hosts_lxc_overlayfs-debian-trixie https://zuul.opendev.org/t/openstack/build/fd1988cd3228457499297ee1f95a8486 : SUCCESS in 22m 33s","accounts_in_message":[],"_revision_number":1},{"id":"31e51a500889f57cd3e0cea76a8a221a3c28b47d","tag":"autogenerated:zuul:gate","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2026-01-22 23:27:52.000000000","message":"Patch Set 1: -Verified\n\nStarting gate jobs.","accounts_in_message":[],"_revision_number":1},{"id":"bb1a778706a21a39c40cc681a75a3551b23acf28","tag":"autogenerated:zuul:gate","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2026-01-23 00:54:58.000000000","message":"Patch Set 1: Verified-2\n\nBuild failed (gate pipeline).  For information on how to proceed, see\nhttps://docs.opendev.org/opendev/infra-manual/latest/developers.html#automated-testing\n\nhttps://zuul.opendev.org/t/openstack/buildset/e0162736ce4243ea9696231883987915\n\n- openstack-ansible-linters-ubuntu-noble https://zuul.opendev.org/t/openstack/build/6ad7e2329b824f16a826185796940781 : SUCCESS in 7m 39s\n- openstack-tox-docs https://zuul.opendev.org/t/openstack/build/3136ca95eb21403ba2209e601fa4e3cf : SUCCESS in 3m 37s\n- openstack-ansible-deploy-infra_lxc-centos-10-stream https://zuul.opendev.org/t/openstack/build/9b7600c70255444491d40c20d1d747f8 : SUCCESS in 45m 33s\n- openstack-ansible-deploy-infra_lxc-debian-bookworm https://zuul.opendev.org/t/openstack/build/6dd998d3e7294c8a8ade5617f2794cdd : SUCCESS in 1h 11m 21s\n- openstack-ansible-deploy-infra_lxc-debian-trixie https://zuul.opendev.org/t/openstack/build/b9938e823e024cb384370095306cfc09 : SUCCESS in 1h 14m 30s\n- openstack-ansible-deploy-infra_lxc-rockylinux-10 https://zuul.opendev.org/t/openstack/build/2bf2a92bc130495ead8858785b337c9c : SUCCESS in 45m 58s\n- openstack-ansible-deploy-infra_lxc_tls-rockylinux-10 https://zuul.opendev.org/t/openstack/build/1403fdc9df5743d2b4f920c16a2b5e3f : SUCCESS in 1h 08m 49s\n- openstack-ansible-deploy-infra_lxc-ubuntu-noble https://zuul.opendev.org/t/openstack/build/cba21e3252af48749f37fad364beb289 : SUCCESS in 1h 16m 48s\n- openstack-ansible-deploy-infra_lxc_stepca-ubuntu-noble https://zuul.opendev.org/t/openstack/build/f14a4715baad4738aa44f5e1be4222cd : SUCCESS in 1h 18m 08s\n- openstack-ansible-deploy-infra_lxc_tls-ubuntu-noble https://zuul.opendev.org/t/openstack/build/c74d1189043b420e9820033d063d54f2 : SUCCESS in 1h 02m 17s\n- openstack-ansible-upgrade-infra_lxc-ubuntu-noble https://zuul.opendev.org/t/openstack/build/ed306ed098564f258577235ed14146a8 : FAILURE in 32m 48s\n- openstack-ansible-deploy-hosts_lxc_zfs-ubuntu-noble https://zuul.opendev.org/t/openstack/build/7e56a17abacd422c8881590b2e8a3e39 : SUCCESS in 38m 29s\n- openstack-ansible-deploy-hosts_lxc_btrfs-ubuntu-noble https://zuul.opendev.org/t/openstack/build/39350d89b2b2488d9b33f562cdf9e32f : SUCCESS in 14m 49s\n- openstack-ansible-deploy-hosts_lxc_overlayfs-ubuntu-noble https://zuul.opendev.org/t/openstack/build/3cdc9382c2264a6d81f13a629e14d3a5 : SUCCESS in 21m 05s\n- openstack-ansible-deploy-hosts_lxc_btrfs-debian-bookworm https://zuul.opendev.org/t/openstack/build/c5794f4b74ec4f90a052fc4f5e71836d : SUCCESS in 22m 13s\n- openstack-ansible-deploy-hosts_lxc_overlayfs-debian-bookworm https://zuul.opendev.org/t/openstack/build/c4a9f87997974c249477a8249841ca90 : SUCCESS in 23m 04s\n- openstack-ansible-deploy-hosts_lxc_btrfs-debian-trixie https://zuul.opendev.org/t/openstack/build/ae3019575a4548fab283e2b8d3a9b2fc : SUCCESS in 22m 22s\n- openstack-ansible-deploy-hosts_lxc_overlayfs-debian-trixie https://zuul.opendev.org/t/openstack/build/4fc47fc655fc4b92a46eeabfaf2cf94f : SUCCESS in 16m 10s\n- openstack-ansible-deploy-hosts_lxc_overlayfs-rockylinux-10 https://zuul.opendev.org/t/openstack/build/c8945920d93e437481a80f213a5552d2 : SUCCESS in 16m 31s","accounts_in_message":[],"_revision_number":1},{"id":"bb4a166164d4468ad1db784451771a13294ebe9f","author":{"_account_id":37598,"name":"Ivan Anfimov","display_name":"Ivan Anfimov","email":"lazekteam@gmail.com","username":"anfimovir"},"date":"2026-01-23 07:02:51.000000000","message":"Patch Set 1:\n\n(1 comment)","accounts_in_message":[],"_revision_number":1},{"id":"ef28864384db9eba2a16469e933209173f0b6194","tag":"autogenerated:zuul:check","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2026-01-23 09:03:36.000000000","message":"Patch Set 1: Verified-1\n\nBuild failed (check pipeline).  For information on how to proceed, see\nhttps://docs.opendev.org/opendev/infra-manual/latest/developers.html#automated-testing\nand https://docs.openstack.org/project-team-guide/testing.html#how-to-handle-test-failures\n\nhttps://zuul.opendev.org/t/openstack/buildset/86cb28be735542e0aa24b7a7c04c12eb\n\n- openstack-ansible-linters-ubuntu-noble https://zuul.opendev.org/t/openstack/build/76745d56571d48a0840078fbed3b1267 : SUCCESS in 5m 32s\n- openstack-tox-docs https://zuul.opendev.org/t/openstack/build/4e3ee60ad67d423685e99688d6069e90 : SUCCESS in 4m 48s\n- openstack-ansible-deploy-infra_lxc-centos-10-stream https://zuul.opendev.org/t/openstack/build/4b60230441a244619b3c71e1d39ff743 : SUCCESS in 45m 17s\n- openstack-ansible-deploy-infra_lxc-debian-bookworm https://zuul.opendev.org/t/openstack/build/abb44775b6a2449387404ce644c1feba : SUCCESS in 1h 20m 27s\n- openstack-ansible-deploy-infra_lxc-debian-trixie https://zuul.opendev.org/t/openstack/build/d17006cd0e7a4cad9be2873cca77d190 : SUCCESS in 1h 17m 34s\n- openstack-ansible-deploy-infra_lxc-rockylinux-10 https://zuul.opendev.org/t/openstack/build/ae3fea784d324bd3a8b87530fd0caf17 : SUCCESS in 45m 17s\n- openstack-ansible-deploy-infra_lxc_tls-rockylinux-10 https://zuul.opendev.org/t/openstack/build/faf3c5b93d76482ea6fb39c3145a574d : SUCCESS in 1h 11m 39s\n- openstack-ansible-deploy-infra_lxc-ubuntu-noble https://zuul.opendev.org/t/openstack/build/e1a4000673fe43d88e1e657f4310f45b : SUCCESS in 1h 02m 09s\n- openstack-ansible-deploy-infra_lxc_stepca-ubuntu-noble https://zuul.opendev.org/t/openstack/build/1b3e41b4250d4da194faf7fe5bba203f : SUCCESS in 1h 05m 03s\n- openstack-ansible-deploy-infra_lxc_tls-ubuntu-noble https://zuul.opendev.org/t/openstack/build/b8ea3599e4594f81adb5c3a7993dfa67 : FAILURE in 1h 26m 16s\n- openstack-ansible-upgrade-infra_lxc-ubuntu-noble https://zuul.opendev.org/t/openstack/build/e9151d09b9ff4830b30bcbb6eba8c31d : SUCCESS in 1h 56m 13s\n- openstack-ansible-deploy-hosts_lxc_zfs-ubuntu-noble https://zuul.opendev.org/t/openstack/build/3f8b3f3dffae45be9053a9308bec922b : SUCCESS in 18m 56s\n- openstack-ansible-deploy-hosts_lxc_btrfs-ubuntu-noble https://zuul.opendev.org/t/openstack/build/ede501b557104d88a33dd5f333d22ff4 : SUCCESS in 23m 17s\n- openstack-ansible-deploy-hosts_lxc_overlayfs-ubuntu-noble https://zuul.opendev.org/t/openstack/build/7ed958006bfd4def821eac87390fab09 : SUCCESS in 25m 41s\n- openstack-ansible-deploy-hosts_lxc_btrfs-debian-bookworm https://zuul.opendev.org/t/openstack/build/d6722af14c33479296776bb355c55a9a : SUCCESS in 29m 05s\n- openstack-ansible-deploy-hosts_lxc_overlayfs-debian-bookworm https://zuul.opendev.org/t/openstack/build/54c3f34575a143cbbd790a298700cef2 : SUCCESS in 15m 08s\n- openstack-ansible-deploy-hosts_lxc_overlayfs-rockylinux-10 https://zuul.opendev.org/t/openstack/build/004ae844f7ce494bbe678914fe2fc0c9 : SUCCESS in 15m 44s\n- openstack-ansible-deploy-hosts_lxc_btrfs-debian-trixie https://zuul.opendev.org/t/openstack/build/c67815c1b12b40409da6f312dbfe8d95 : SUCCESS in 21m 39s\n- openstack-ansible-deploy-hosts_lxc_overlayfs-debian-trixie https://zuul.opendev.org/t/openstack/build/3748292842af430ab06f1f0e740ce493 : SUCCESS in 22m 08s","accounts_in_message":[],"_revision_number":1},{"id":"306ae48bc9cdf20cdafd64992ab70d959b344fd2","author":{"_account_id":37598,"name":"Ivan Anfimov","display_name":"Ivan Anfimov","email":"lazekteam@gmail.com","username":"anfimovir"},"date":"2026-01-23 13:51:06.000000000","message":"Patch Set 1:\n\n(1 comment)","accounts_in_message":[],"_revision_number":1},{"id":"e04bdbd3e6001f78b61010f988e345cd75a3dcc2","tag":"autogenerated:zuul:check","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2026-01-23 15:44:01.000000000","message":"Patch Set 1: Verified+1\n\nBuild succeeded (check pipeline).\nhttps://zuul.opendev.org/t/openstack/buildset/e8894a3ace6541e3aab9f3d9fa473eab\n\n- openstack-ansible-linters-ubuntu-noble https://zuul.opendev.org/t/openstack/build/93e34a4906ca41d3bcd4a3037f3e9abe : SUCCESS in 8m 08s\n- openstack-tox-docs https://zuul.opendev.org/t/openstack/build/f96f9855501a4746930dc2cad6b3d8c7 : SUCCESS in 5m 23s\n- openstack-ansible-deploy-infra_lxc-centos-10-stream https://zuul.opendev.org/t/openstack/build/4af02b40b8de433b9c6827ef1101fbd6 : SUCCESS in 1h 08m 27s\n- openstack-ansible-deploy-infra_lxc-debian-bookworm https://zuul.opendev.org/t/openstack/build/cd63e80bb55846f49a017ff4700d0286 : SUCCESS in 1h 16m 38s\n- openstack-ansible-deploy-infra_lxc-debian-trixie https://zuul.opendev.org/t/openstack/build/386e61e1b1e14c1094d12a4931d1dc0c : SUCCESS in 1h 05m 48s\n- openstack-ansible-deploy-infra_lxc-rockylinux-10 https://zuul.opendev.org/t/openstack/build/b527eb5e70db4b368622816b89a84b43 : SUCCESS in 1h 14m 37s\n- openstack-ansible-deploy-infra_lxc_tls-rockylinux-10 https://zuul.opendev.org/t/openstack/build/60a133f297df4124a2c956eeb713229f : SUCCESS in 1h 05m 42s\n- openstack-ansible-deploy-infra_lxc-ubuntu-noble https://zuul.opendev.org/t/openstack/build/f65dcb88aa8f4dbca5b257e618778689 : SUCCESS in 59m 35s\n- openstack-ansible-deploy-infra_lxc_stepca-ubuntu-noble https://zuul.opendev.org/t/openstack/build/5e8047ec013f432db93e9c1c92209343 : SUCCESS in 1h 05m 06s\n- openstack-ansible-deploy-infra_lxc_tls-ubuntu-noble https://zuul.opendev.org/t/openstack/build/5525746751d94b9c9ffb3f0cd978fdda : SUCCESS in 1h 00m 12s\n- openstack-ansible-upgrade-infra_lxc-ubuntu-noble https://zuul.opendev.org/t/openstack/build/a7e02bd550224530aa25afed7e44cc87 : SUCCESS in 1h 45m 00s\n- openstack-ansible-deploy-hosts_lxc_zfs-ubuntu-noble https://zuul.opendev.org/t/openstack/build/8667094007254f57b9fb3e63650ae2ae : SUCCESS in 25m 21s\n- openstack-ansible-deploy-hosts_lxc_btrfs-ubuntu-noble https://zuul.opendev.org/t/openstack/build/c4aa049fa9a74ceabd7f872663ecf78b : SUCCESS in 39m 55s\n- openstack-ansible-deploy-hosts_lxc_overlayfs-ubuntu-noble https://zuul.opendev.org/t/openstack/build/48d024d4a72241068f2ed555e8f285e9 : SUCCESS in 35m 28s\n- openstack-ansible-deploy-hosts_lxc_btrfs-debian-bookworm https://zuul.opendev.org/t/openstack/build/3063a12dada44f779ec19596fdab9769 : SUCCESS in 25m 09s\n- openstack-ansible-deploy-hosts_lxc_overlayfs-debian-bookworm https://zuul.opendev.org/t/openstack/build/d02e0b516af14987a6a2ab977752b479 : SUCCESS in 21m 38s\n- openstack-ansible-deploy-hosts_lxc_overlayfs-rockylinux-10 https://zuul.opendev.org/t/openstack/build/ea06d42471a647bda202ebf5bc9292f2 : SUCCESS in 15m 42s\n- openstack-ansible-deploy-hosts_lxc_btrfs-debian-trixie https://zuul.opendev.org/t/openstack/build/d4e8f04f3f914d66ac869d4b0852a47f : SUCCESS in 23m 48s\n- openstack-ansible-deploy-hosts_lxc_overlayfs-debian-trixie https://zuul.opendev.org/t/openstack/build/d38e1fe1230b4ad987768a2c8e3791df : SUCCESS in 18m 14s","accounts_in_message":[],"_revision_number":1},{"id":"2e98eba64354b98878e425f6a0b4dea4c0b069da","tag":"autogenerated:zuul:gate","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2026-01-23 15:46:21.000000000","message":"Patch Set 1: -Verified\n\nStarting gate jobs.","accounts_in_message":[],"_revision_number":1},{"id":"7dcbd3a7f495c9f00675820d5b6d17bc8568ae2a","tag":"autogenerated:zuul:gate","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2026-01-23 17:25:52.000000000","message":"Patch Set 1: Verified+2\n\nBuild succeeded (gate pipeline).\nhttps://zuul.opendev.org/t/openstack/buildset/7eaae1bf19e2435488370286596f3f77\n\n- openstack-ansible-linters-ubuntu-noble https://zuul.opendev.org/t/openstack/build/eb3afe629ba646719112048097eae7ff : SUCCESS in 12m 23s\n- openstack-tox-docs https://zuul.opendev.org/t/openstack/build/9cdb18371b6547b0853382093e5a4ae8 : SUCCESS in 6m 10s\n- openstack-ansible-deploy-infra_lxc-centos-10-stream https://zuul.opendev.org/t/openstack/build/4224cd7e02fc4c32b0759c24c5b3fdd0 : SUCCESS in 47m 39s\n- openstack-ansible-deploy-infra_lxc-debian-bookworm https://zuul.opendev.org/t/openstack/build/f9c1743432dd460090d920c9952d2a07 : SUCCESS in 45m 54s\n- openstack-ansible-deploy-infra_lxc-debian-trixie https://zuul.opendev.org/t/openstack/build/f62831065a0f4e0aa1e8498f48ca7d66 : SUCCESS in 1h 31m 18s\n- openstack-ansible-deploy-infra_lxc-rockylinux-10 https://zuul.opendev.org/t/openstack/build/3f7f2dfb687f48ddaa64b9ffa5d7d315 : SUCCESS in 1h 18m 51s\n- openstack-ansible-deploy-infra_lxc_tls-rockylinux-10 https://zuul.opendev.org/t/openstack/build/b2eeb376723f4171ac6bd2b14a362e64 : SUCCESS in 49m 43s\n- openstack-ansible-deploy-infra_lxc-ubuntu-noble https://zuul.opendev.org/t/openstack/build/f61378d6634e4395aae1809c8a1ebf00 : SUCCESS in 1h 20m 09s\n- openstack-ansible-deploy-infra_lxc_stepca-ubuntu-noble https://zuul.opendev.org/t/openstack/build/27408e4db2934d5799830a4ae9f46730 : SUCCESS in 1h 12m 24s\n- openstack-ansible-deploy-infra_lxc_tls-ubuntu-noble https://zuul.opendev.org/t/openstack/build/c2eb19497e7b4c8091cb444896623ac7 : SUCCESS in 1h 14m 06s\n- openstack-ansible-upgrade-infra_lxc-ubuntu-noble https://zuul.opendev.org/t/openstack/build/b5baf31245f44a38a01995a17e936f71 : SUCCESS in 56m 06s\n- openstack-ansible-deploy-hosts_lxc_zfs-ubuntu-noble https://zuul.opendev.org/t/openstack/build/6ff5337f8b1b42e297d25f3472b642a1 : SUCCESS in 29m 55s\n- openstack-ansible-deploy-hosts_lxc_btrfs-ubuntu-noble https://zuul.opendev.org/t/openstack/build/ab7bd9e50da542688266ab659e6dddc1 : SUCCESS in 30m 47s\n- openstack-ansible-deploy-hosts_lxc_overlayfs-ubuntu-noble https://zuul.opendev.org/t/openstack/build/173a85a6d46d46c79baf34655ac62e7d : SUCCESS in 32m 00s\n- openstack-ansible-deploy-hosts_lxc_btrfs-debian-bookworm https://zuul.opendev.org/t/openstack/build/5394e5e49f4947eaba1f70bfb0697bb4 : SUCCESS in 26m 44s\n- openstack-ansible-deploy-hosts_lxc_overlayfs-debian-bookworm https://zuul.opendev.org/t/openstack/build/c7e8ccd8877a4838b52d69c28424f0bf : SUCCESS in 27m 06s\n- openstack-ansible-deploy-hosts_lxc_btrfs-debian-trixie https://zuul.opendev.org/t/openstack/build/94bba7ce1f224518871051a2aa887483 : SUCCESS in 27m 31s\n- openstack-ansible-deploy-hosts_lxc_overlayfs-debian-trixie https://zuul.opendev.org/t/openstack/build/2bfb956b3c9c4476ab63a1a993219167 : SUCCESS in 26m 45s\n- openstack-ansible-deploy-hosts_lxc_overlayfs-rockylinux-10 https://zuul.opendev.org/t/openstack/build/472f52220c7e4aac83b110c129615a37 : SUCCESS in 26m 01s","accounts_in_message":[],"_revision_number":1},{"id":"a25a0b9d2ea20f66ee3ddadf5d7cc37aed6e53d9","tag":"autogenerated:gerrit:merged","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2026-01-23 17:25:52.000000000","message":"Change has been successfully merged","accounts_in_message":[],"_revision_number":1},{"id":"9cdaa2287f4c589770946cb28dfddebc73ee8a0d","tag":"autogenerated:zuul:promote","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2026-01-23 17:27:50.000000000","message":"Patch Set 1:\n\nBuild succeeded (promote pipeline).\nhttps://zuul.opendev.org/t/openstack/buildset/2b92f161bf52412fab70cfcc15d8b81d\n\n- promote-openstack-tox-docs https://zuul.opendev.org/t/openstack/build/4b1de21281724804851d2977b056378a : SUCCESS in 43s","accounts_in_message":[],"_revision_number":1}],"current_revision_number":1,"current_revision":"fce682be8002d25af3f7ec7ca35e424ae22a442e","revisions":{"fce682be8002d25af3f7ec7ca35e424ae22a442e":{"kind":"REWORK","_number":1,"created":"2026-01-09 11:44:15.000000000","uploader":{"_account_id":28619,"name":"Dmitriy Rabotyagov","email":"noonedeadpunk@gmail.com","username":"noonedeadpunk"},"ref":"refs/changes/48/972848/1","fetch":{"anonymous http":{"url":"https://review.opendev.org/openstack/openstack-ansible-lxc_hosts","ref":"refs/changes/48/972848/1","commands":{"Checkout":"git fetch https://review.opendev.org/openstack/openstack-ansible-lxc_hosts refs/changes/48/972848/1 \u0026\u0026 git checkout FETCH_HEAD","Cherry Pick":"git fetch https://review.opendev.org/openstack/openstack-ansible-lxc_hosts refs/changes/48/972848/1 \u0026\u0026 git cherry-pick FETCH_HEAD","Format Patch":"git fetch https://review.opendev.org/openstack/openstack-ansible-lxc_hosts refs/changes/48/972848/1 \u0026\u0026 git format-patch -1 --stdout FETCH_HEAD","Pull":"git pull https://review.opendev.org/openstack/openstack-ansible-lxc_hosts refs/changes/48/972848/1"}}},"commit":{"parents":[{"commit":"2e07ee099268fa0d55f0e1b837aac8a3fe17ae4b","subject":"Do not install post-quantum GPG keys","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/openstack-ansible-lxc_hosts/commit/2e07ee099268fa0d55f0e1b837aac8a3fe17ae4b"}]}],"author":{"name":"Dmitriy Rabotyagov","email":"dmitriy.rabotyagov@cleura.com","date":"2026-01-09 11:42:49.000000000","tz":60},"committer":{"name":"Dmitriy Rabotyagov","email":"dmitriy.rabotyagov@cleura.com","date":"2026-01-09 11:44:08.000000000","tz":60},"subject":"Revert \"Temporary disable RHEL jobs\"","message":"Revert \"Temporary disable RHEL jobs\"\n\nThis reverts commit f9fbada7eada6de490577c1af56195f6ea9c5789.\n\nDepends-On: https://review.opendev.org/c/openstack/openstack-ansible-rabbitmq_server/+/972497\nChange-Id: Idccfd72ce57e5fdd0f7ad6e2def5dd153f8ddbbe\nSigned-off-by: Dmitriy Rabotyagov \u003cdmitriy.rabotyagov@cleura.com\u003e\n","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/openstack-ansible-lxc_hosts/commit/fce682be8002d25af3f7ec7ca35e424ae22a442e"}],"resolve_conflicts_web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/openstack-ansible-lxc_hosts/commit/fce682be8002d25af3f7ec7ca35e424ae22a442e"}]},"branch":"refs/heads/master"}},"requirements":[],"submit_records":[{"rule_name":"gerrit~DefaultSubmitRule","status":"CLOSED","labels":[{"label":"Verified","status":"MAY","applied_by":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]}},{"label":"Code-Review","status":"MAY","applied_by":{"_account_id":37632,"name":"Dmitriy Chubinidze","email":"dcu995@gmail.com","username":"chubinidzedr"}},{"label":"Workflow","status":"MAY","applied_by":{"_account_id":28619,"name":"Dmitriy Rabotyagov","email":"noonedeadpunk@gmail.com","username":"noonedeadpunk"}},{"label":"Backport-Candidate","status":"MAY"}]}],"submit_requirements":[{"name":"Verified","description":"Verified in gate by CI","status":"SATISFIED","is_legacy":false,"submittability_expression_result":{"expression":"label:Verified\u003dMAX AND -label:Verified\u003dMIN","fulfilled":true,"status":"PASS","passing_atoms":["label:Verified\u003dMAX"],"failing_atoms":["label:Verified\u003dMIN"],"atom_explanations":{}}},{"name":"Backport-Candidate","description":"Backport candidate status","status":"NOT_APPLICABLE","is_legacy":false,"applicability_expression_result":{"fulfilled":false,"status":"FAIL"},"submittability_expression_result":{"expression":"is:true","fulfilled":true,"status":"NOT_EVALUATED","passing_atoms":[],"failing_atoms":[],"atom_explanations":{}}},{"name":"Code-Review","description":"Code reviewed by core reviewer","status":"SATISFIED","is_legacy":false,"submittability_expression_result":{"expression":"label:Code-Review\u003dMAX AND -label:Code-Review\u003dMIN","fulfilled":true,"status":"PASS","passing_atoms":["label:Code-Review\u003dMAX"],"failing_atoms":["label:Code-Review\u003dMIN"],"atom_explanations":{}}},{"name":"Workflow","description":"Approved for gate by core reviewer","status":"SATISFIED","is_legacy":false,"submittability_expression_result":{"expression":"label:Workflow\u003dMAX AND -label:Workflow\u003dMIN","fulfilled":true,"status":"PASS","passing_atoms":["label:Workflow\u003dMAX"],"failing_atoms":["label:Workflow\u003dMIN"],"atom_explanations":{}}}]}
