)]}'
{"id":"openstack%2Fopenstack-ansible-lxc_hosts~972870","triplet_id":"openstack%2Fopenstack-ansible-lxc_hosts~stable%2F2025.1~If176d905411869b55ff1751659f5e46963c9f037","project":"openstack/openstack-ansible-lxc_hosts","branch":"stable/2025.1","attention_set":{},"removed_from_attention_set":{"31542":{"account":{"_account_id":31542,"name":"Andrew Bonney","email":"andrew.bonney@bbc.co.uk","username":"andrewbonney"},"last_update":"2026-01-19 11:02:55.000000000","reason":"\u003cGERRIT_ACCOUNT_31542\u003e replied on the change","reason_account":{"_account_id":31542,"name":"Andrew Bonney","email":"andrew.bonney@bbc.co.uk","username":"andrewbonney"}},"32666":{"account":{"_account_id":32666,"name":"Damian Dąbrowski","email":"damian@dabrowski.cloud","username":"ddabrowski"},"last_update":"2026-01-19 13:31:08.000000000","reason":"Change was submitted"},"28619":{"account":{"_account_id":28619,"name":"Dmitriy Rabotyagov","email":"noonedeadpunk@gmail.com","username":"noonedeadpunk"},"last_update":"2026-01-19 11:10:41.000000000","reason":"\u003cGERRIT_ACCOUNT_28619\u003e replied on the change","reason_account":{"_account_id":28619,"name":"Dmitriy Rabotyagov","email":"noonedeadpunk@gmail.com","username":"noonedeadpunk"}},"37598":{"account":{"_account_id":37598,"name":"Ivan Anfimov","display_name":"Ivan Anfimov","email":"lazekteam@gmail.com","username":"anfimovir"},"last_update":"2026-01-19 13:31:08.000000000","reason":"Change was submitted"},"25023":{"account":{"_account_id":25023,"name":"Jonathan Rosser","email":"jonathan.rosser@rd.bbc.co.uk","username":"jrosser"},"last_update":"2026-01-19 13:31:08.000000000","reason":"Change was submitted"}},"hashtags":[],"change_id":"If176d905411869b55ff1751659f5e46963c9f037","subject":"Do not install post-quantum GPG keys","status":"MERGED","created":"2026-01-09 13:39:39.000000000","updated":"2026-01-19 13:33:53.000000000","submitted":"2026-01-19 13:31:08.000000000","submitter":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"total_comment_count":4,"unresolved_comment_count":0,"has_review_started":true,"submission_id":"972870","cherry_pick_of_change":972599,"cherry_pick_of_patch_set":2,"meta_rev_id":"30624bea89164d9a5eb9c90fdab212adf98e60ea","_number":972870,"virtual_id_number":972870,"owner":{"_account_id":37598,"name":"Ivan Anfimov","display_name":"Ivan Anfimov","email":"lazekteam@gmail.com","username":"anfimovir"},"actions":{},"labels":{"Verified":{"approved":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"all":[{"value":0,"_account_id":31542,"name":"Andrew Bonney","email":"andrew.bonney@bbc.co.uk","username":"andrewbonney"},{"value":0,"_account_id":28619,"name":"Dmitriy Rabotyagov","email":"noonedeadpunk@gmail.com","username":"noonedeadpunk"},{"tag":"autogenerated:zuul:gate","value":2,"date":"2026-01-19 13:31:07.000000000","permitted_voting_range":{"min":2,"max":2},"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},{"value":0,"_account_id":37598,"name":"Ivan Anfimov","display_name":"Ivan Anfimov","email":"lazekteam@gmail.com","username":"anfimovir"}],"values":{"-2":"Fails","-1":"Doesn\u0027t seem to work"," 0":"No score","+1":"Works for me","+2":"Verified"},"description":"","default_value":0,"optional":true},"Code-Review":{"approved":{"_account_id":31542,"name":"Andrew Bonney","email":"andrew.bonney@bbc.co.uk","username":"andrewbonney"},"all":[{"value":2,"date":"2026-01-19 11:02:55.000000000","permitted_voting_range":{"min":2,"max":2},"_account_id":31542,"name":"Andrew Bonney","email":"andrew.bonney@bbc.co.uk","username":"andrewbonney"},{"value":2,"date":"2026-01-19 11:10:41.000000000","permitted_voting_range":{"min":2,"max":2},"_account_id":28619,"name":"Dmitriy Rabotyagov","email":"noonedeadpunk@gmail.com","username":"noonedeadpunk"},{"value":0,"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},{"value":1,"date":"2026-01-19 10:30:59.000000000","permitted_voting_range":{"min":1,"max":2},"_account_id":37598,"name":"Ivan Anfimov","display_name":"Ivan Anfimov","email":"lazekteam@gmail.com","username":"anfimovir"}],"values":{"-2":"Do not merge","-1":"This patch needs further work before it can be merged"," 0":"No score","+1":"Looks good to me, but someone else must approve","+2":"Looks good to me (core reviewer)"},"description":"","default_value":0,"optional":true},"Workflow":{"approved":{"_account_id":28619,"name":"Dmitriy Rabotyagov","email":"noonedeadpunk@gmail.com","username":"noonedeadpunk"},"all":[{"value":0,"_account_id":31542,"name":"Andrew Bonney","email":"andrew.bonney@bbc.co.uk","username":"andrewbonney"},{"value":1,"date":"2026-01-19 11:10:41.000000000","permitted_voting_range":{"min":1,"max":1},"_account_id":28619,"name":"Dmitriy Rabotyagov","email":"noonedeadpunk@gmail.com","username":"noonedeadpunk"},{"value":0,"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},{"value":0,"_account_id":37598,"name":"Ivan Anfimov","display_name":"Ivan Anfimov","email":"lazekteam@gmail.com","username":"anfimovir"}],"values":{"-1":"Work in progress"," 0":"Ready for reviews","+1":"Approved"},"description":"","default_value":0,"optional":true},"Backport-Candidate":{"all":[{"value":0,"_account_id":31542,"name":"Andrew Bonney","email":"andrew.bonney@bbc.co.uk","username":"andrewbonney"},{"value":0,"_account_id":28619,"name":"Dmitriy Rabotyagov","email":"noonedeadpunk@gmail.com","username":"noonedeadpunk"},{"value":0,"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},{"value":0,"_account_id":37598,"name":"Ivan Anfimov","display_name":"Ivan Anfimov","email":"lazekteam@gmail.com","username":"anfimovir"}],"values":{"-1":"Not A Backport Candidate"," 0":"Backport Review Needed","+1":"Proposed Backport"},"description":"","default_value":0,"optional":true}},"removable_reviewers":[{"_account_id":32666,"name":"Damian Dąbrowski","email":"damian@dabrowski.cloud","username":"ddabrowski"},{"_account_id":25023,"name":"Jonathan Rosser","email":"jonathan.rosser@rd.bbc.co.uk","username":"jrosser"}],"reviewers":{"REVIEWER":[{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},{"_account_id":25023,"name":"Jonathan Rosser","email":"jonathan.rosser@rd.bbc.co.uk","username":"jrosser"},{"_account_id":28619,"name":"Dmitriy Rabotyagov","email":"noonedeadpunk@gmail.com","username":"noonedeadpunk"},{"_account_id":31542,"name":"Andrew Bonney","email":"andrew.bonney@bbc.co.uk","username":"andrewbonney"},{"_account_id":32666,"name":"Damian Dąbrowski","email":"damian@dabrowski.cloud","username":"ddabrowski"},{"_account_id":37598,"name":"Ivan Anfimov","display_name":"Ivan Anfimov","email":"lazekteam@gmail.com","username":"anfimovir"}]},"pending_reviewers":{},"reviewer_updates":[{"updated":"2026-01-09 13:39:39.000000000","updated_by":{"_account_id":37598,"name":"Ivan Anfimov","display_name":"Ivan Anfimov","email":"lazekteam@gmail.com","username":"anfimovir"},"reviewer":{"_account_id":28619,"name":"Dmitriy Rabotyagov","email":"noonedeadpunk@gmail.com","username":"noonedeadpunk"},"state":"CC"},{"updated":"2026-01-09 13:45:32.000000000","updated_by":{"_account_id":37598,"name":"Ivan Anfimov","display_name":"Ivan Anfimov","email":"lazekteam@gmail.com","username":"anfimovir"},"reviewer":{"_account_id":28619,"name":"Dmitriy Rabotyagov","email":"noonedeadpunk@gmail.com","username":"noonedeadpunk"},"state":"REVIEWER"},{"updated":"2026-01-09 13:45:32.000000000","updated_by":{"_account_id":37598,"name":"Ivan Anfimov","display_name":"Ivan Anfimov","email":"lazekteam@gmail.com","username":"anfimovir"},"reviewer":{"_account_id":31542,"name":"Andrew Bonney","email":"andrew.bonney@bbc.co.uk","username":"andrewbonney"},"state":"REVIEWER"},{"updated":"2026-01-09 13:45:32.000000000","updated_by":{"_account_id":37598,"name":"Ivan Anfimov","display_name":"Ivan Anfimov","email":"lazekteam@gmail.com","username":"anfimovir"},"reviewer":{"_account_id":25023,"name":"Jonathan Rosser","email":"jonathan.rosser@rd.bbc.co.uk","username":"jrosser"},"state":"REVIEWER"},{"updated":"2026-01-09 15:40:38.000000000","updated_by":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"reviewer":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"state":"REVIEWER"},{"updated":"2026-01-09 19:46:05.000000000","updated_by":{"_account_id":37598,"name":"Ivan Anfimov","display_name":"Ivan Anfimov","email":"lazekteam@gmail.com","username":"anfimovir"},"reviewer":{"_account_id":32666,"name":"Damian Dąbrowski","email":"damian@dabrowski.cloud","username":"ddabrowski"},"state":"REVIEWER"}],"messages":[{"id":"7a8bdb7db96bd606a78300c943b60a978f1939c8","tag":"autogenerated:gerrit:newPatchSet","author":{"_account_id":37598,"name":"Ivan Anfimov","display_name":"Ivan Anfimov","email":"lazekteam@gmail.com","username":"anfimovir"},"date":"2026-01-09 13:39:39.000000000","message":"Patch Set 1: Cherry Picked from branch master.","accounts_in_message":[],"_revision_number":1},{"id":"3b6135fa71ef340f6935edb708689fe619da235a","tag":"autogenerated:gerrit:newPatchSet","author":{"_account_id":37598,"name":"Ivan Anfimov","display_name":"Ivan Anfimov","email":"lazekteam@gmail.com","username":"anfimovir"},"date":"2026-01-09 13:43:03.000000000","message":"Patch Set 2: Patch Set 1 was rebased","accounts_in_message":[],"_revision_number":2},{"id":"bc1438a74b28eaacc53c318a230a7d5ad8a6fec1","tag":"autogenerated:zuul:check","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2026-01-09 15:40:38.000000000","message":"Patch Set 2: Verified-1\n\nBuild failed (check pipeline).  For information on how to proceed, see\nhttps://docs.opendev.org/opendev/infra-manual/latest/developers.html#automated-testing\nand https://docs.openstack.org/project-team-guide/testing.html#how-to-handle-test-failures\n\nhttps://zuul.opendev.org/t/openstack/buildset/f825e3d9dd874e2bb49737f898d94a24\n\n- openstack-ansible-linters-ubuntu-noble https://zuul.opendev.org/t/openstack/build/5b139f44c442440d818c1583fc5ce297 : SUCCESS in 11m 26s\n- openstack-tox-docs https://zuul.opendev.org/t/openstack/build/36afef99e1f54e11b845682c1776321f : SUCCESS in 7m 25s\n- openstack-ansible-deploy-infra_lxc-centos-9-stream https://zuul.opendev.org/t/openstack/build/8dcc31c6fbe946b7833955e445ca7c29 : FAILURE in 52m 39s (non-voting)\n- openstack-ansible-deploy-infra_lxc-rockylinux-9 https://zuul.opendev.org/t/openstack/build/bb25b6cb61274443aae98da616acc2d9 : FAILURE in 51m 15s\n- openstack-ansible-deploy-infra_lxc_tls-rockylinux-9 https://zuul.opendev.org/t/openstack/build/505e019e7d3e443793e5f1e3a039adfc : FAILURE in 52m 24s\n- openstack-ansible-deploy-infra_lxc-rockylinux-10 https://zuul.opendev.org/t/openstack/build/634c913a353a485f9cc7334e9df41da1 : FAILURE in 43m 30s (non-voting)\n- openstack-ansible-deploy-infra_lxc_tls-rockylinux-10 https://zuul.opendev.org/t/openstack/build/b16d1ef4af024ff3a18aae2f7935e1c4 : FAILURE in 41m 42s (non-voting)\n- openstack-ansible-deploy-infra_lxc-centos-10-stream https://zuul.opendev.org/t/openstack/build/b77b4f0cce154ae394f781c622157894 : FAILURE in 43m 07s (non-voting)\n- openstack-ansible-deploy-infra_lxc-debian-bookworm https://zuul.opendev.org/t/openstack/build/cfef275bc7cd41f7b4aa33265d11bddb : SUCCESS in 1h 14m 09s\n- openstack-ansible-deploy-infra_lxc-ubuntu-jammy https://zuul.opendev.org/t/openstack/build/acea9252c6b545dc92788271d853a153 : SUCCESS in 1h 17m 46s\n- openstack-ansible-deploy-infra_lxc_stepca-ubuntu-jammy https://zuul.opendev.org/t/openstack/build/66708de9c6944653917268ca70853c9a : SUCCESS in 1h 07m 37s\n- openstack-ansible-deploy-infra_lxc_tls-ubuntu-jammy https://zuul.opendev.org/t/openstack/build/0f7305a8efe74096a55fe73ff38354b8 : SUCCESS in 1h 15m 44s\n- openstack-ansible-deploy-infra_lxc-ubuntu-noble https://zuul.opendev.org/t/openstack/build/7206c8c5944e44f0ad2a658908a2eb62 : SUCCESS in 1h 28m 04s\n- openstack-ansible-deploy-infra_lxc_stepca-ubuntu-noble https://zuul.opendev.org/t/openstack/build/25223d21d23d4012a5bfaa19cbd32ab7 : POST_FAILURE in 1h 19m 31s\n- openstack-ansible-deploy-infra_lxc_tls-ubuntu-noble https://zuul.opendev.org/t/openstack/build/d00f9abd0c6b4e2caac93c8002088867 : SUCCESS in 57m 37s\n- openstack-ansible-upgrade-infra_lxc-rockylinux-9 https://zuul.opendev.org/t/openstack/build/83646559c094440fa9c3c0d20826007f : FAILURE in 44m 08s\n- openstack-ansible-upgrade_2024.1-infra_lxc-rockylinux-9 https://zuul.opendev.org/t/openstack/build/3abab650231d42c68c4ac2f70c39298d : FAILURE in 36m 28s\n- openstack-ansible-upgrade-infra_lxc-ubuntu-jammy https://zuul.opendev.org/t/openstack/build/db95224fee7f4b1a970affd52c000fdb : SUCCESS in 1h 29m 25s\n- openstack-ansible-upgrade_2024.1-infra_lxc-ubuntu-jammy https://zuul.opendev.org/t/openstack/build/72f8c9866ace43eab1fbe52da42d7326 : FAILURE in 27m 43s\n- openstack-ansible-deploy-hosts_lxc_zfs-ubuntu-jammy https://zuul.opendev.org/t/openstack/build/507c800c944740328cd0b682ad3247d4 : SUCCESS in 34m 02s\n- openstack-ansible-deploy-hosts_lxc_btrfs-ubuntu-jammy https://zuul.opendev.org/t/openstack/build/06d57f6a251446bea46820d22bab0c1f : SUCCESS in 31m 37s\n- openstack-ansible-deploy-hosts_lxc_overlayfs-ubuntu-jammy https://zuul.opendev.org/t/openstack/build/57ba48c8e0c74347962be523a26ceea6 : SUCCESS in 27m 58s\n- openstack-ansible-deploy-hosts_lxc_zfs-ubuntu-noble https://zuul.opendev.org/t/openstack/build/4e8ac92f4b074e2792d686435e9c8763 : SUCCESS in 29m 44s\n- openstack-ansible-deploy-hosts_lxc_btrfs-ubuntu-noble https://zuul.opendev.org/t/openstack/build/a97d178b10254f15bedd4c4017a40bf0 : SUCCESS in 26m 49s\n- openstack-ansible-deploy-hosts_lxc_overlayfs-ubuntu-noble https://zuul.opendev.org/t/openstack/build/821c9fc8ba2f4e188954ad8c60dfc92a : SUCCESS in 29m 14s\n- openstack-ansible-deploy-hosts_lxc_btrfs-debian-bookworm https://zuul.opendev.org/t/openstack/build/daff25d05f314c28a29d801ac6d30b0f : SUCCESS in 28m 19s\n- openstack-ansible-deploy-hosts_lxc_overlayfs-debian-bookworm https://zuul.opendev.org/t/openstack/build/1fe5092a02de4a83a36bf876cc1f16f8 : SUCCESS in 25m 14s\n- openstack-ansible-deploy-hosts_lxc_overlayfs-rockylinux-9 https://zuul.opendev.org/t/openstack/build/b4c3995e5993452492c8efeabcf8c3df : SUCCESS in 20m 39s\n- openstack-ansible-deploy-hosts_lxc_overlayfs-rockylinux-10 https://zuul.opendev.org/t/openstack/build/bf1a1cbc6eab49668eb59ec0ea3d2c68 : SUCCESS in 33m 14s","accounts_in_message":[],"_revision_number":2},{"id":"da24ba971cfc0aabcc176ae2cec49b94c8859e1d","author":{"_account_id":37598,"name":"Ivan Anfimov","display_name":"Ivan Anfimov","email":"lazekteam@gmail.com","username":"anfimovir"},"date":"2026-01-09 15:44:30.000000000","message":"Patch Set 2:\n\n(1 comment)","accounts_in_message":[],"_revision_number":2},{"id":"f3d427d64a268850c52c3155b50d8b3b3d847536","tag":"autogenerated:gerrit:newPatchSet","author":{"_account_id":37598,"name":"Ivan Anfimov","display_name":"Ivan Anfimov","email":"lazekteam@gmail.com","username":"anfimovir"},"date":"2026-01-09 15:44:34.000000000","message":"Patch Set 3: Patch Set 2 was rebased\n\nOutdated Votes:\n* Verified-1\n","accounts_in_message":[],"_revision_number":3},{"id":"d91c778f1894607a1493fd8ce4fbe7b99e66bf11","tag":"autogenerated:zuul:check","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2026-01-09 17:35:35.000000000","message":"Patch Set 3: Verified-1\n\nBuild failed (check pipeline).  For information on how to proceed, see\nhttps://docs.opendev.org/opendev/infra-manual/latest/developers.html#automated-testing\nand https://docs.openstack.org/project-team-guide/testing.html#how-to-handle-test-failures\n\nhttps://zuul.opendev.org/t/openstack/buildset/0700dd63bba4421182297b6e72cf7e6c\n\n- openstack-ansible-linters-ubuntu-noble https://zuul.opendev.org/t/openstack/build/6654b1ca7de4425cbd4ad3f1e183e097 : SUCCESS in 13m 05s\n- openstack-tox-docs https://zuul.opendev.org/t/openstack/build/4e8c95d4b1bb4dc78577ba6f106bf0d9 : SUCCESS in 4m 40s\n- openstack-ansible-deploy-infra_lxc-centos-9-stream https://zuul.opendev.org/t/openstack/build/20c2955ffbd94f51b73d0c5c61dac596 : FAILURE in 52m 41s (non-voting)\n- openstack-ansible-deploy-infra_lxc-rockylinux-9 https://zuul.opendev.org/t/openstack/build/887279c385004c1e881e15b583a3d5bc : FAILURE in 34m 51s (non-voting)\n- openstack-ansible-deploy-infra_lxc_tls-rockylinux-9 https://zuul.opendev.org/t/openstack/build/48646d45bd05487a981242879818e6f3 : FAILURE in 37m 42s (non-voting)\n- openstack-ansible-deploy-infra_lxc-rockylinux-10 https://zuul.opendev.org/t/openstack/build/a065608adc004fe88340662be643e23b : FAILURE in 42m 09s\n- openstack-ansible-deploy-infra_lxc_tls-rockylinux-10 https://zuul.opendev.org/t/openstack/build/384ef38a48fa41138586237a41996163 : FAILURE in 53m 07s\n- openstack-ansible-deploy-infra_lxc-centos-10-stream https://zuul.opendev.org/t/openstack/build/e5bb662c1c864a09b65235240b01a8f8 : FAILURE in 49m 16s\n- openstack-ansible-deploy-infra_lxc-debian-bookworm https://zuul.opendev.org/t/openstack/build/532aab0262a14665934944c25e66303e : SUCCESS in 1h 15m 49s\n- openstack-ansible-deploy-infra_lxc-ubuntu-jammy https://zuul.opendev.org/t/openstack/build/e9bea783f0ba4bf09e0bc0398fa233a0 : SUCCESS in 50m 43s\n- openstack-ansible-deploy-infra_lxc_stepca-ubuntu-jammy https://zuul.opendev.org/t/openstack/build/ce6d861b6d9a49deb2649d22c27f4f2c : SUCCESS in 1h 24m 12s\n- openstack-ansible-deploy-infra_lxc_tls-ubuntu-jammy https://zuul.opendev.org/t/openstack/build/542533a1e8774520892fce522f5cd7bf : SUCCESS in 1h 07m 39s\n- openstack-ansible-deploy-infra_lxc-ubuntu-noble https://zuul.opendev.org/t/openstack/build/cd1acde7991f49cc94f6f0f459974748 : SUCCESS in 1h 28m 37s\n- openstack-ansible-deploy-infra_lxc_stepca-ubuntu-noble https://zuul.opendev.org/t/openstack/build/65447ef0c92e4246ad239f7a6d5dfa39 : SUCCESS in 1h 38m 26s\n- openstack-ansible-deploy-infra_lxc_tls-ubuntu-noble https://zuul.opendev.org/t/openstack/build/fed3e982a91e47a1b0cd228634479ca2 : SUCCESS in 43m 30s\n- openstack-ansible-upgrade-infra_lxc-rockylinux-9 https://zuul.opendev.org/t/openstack/build/666e621dc17848b89c5417ce22243383 : FAILURE in 50m 56s (non-voting)\n- openstack-ansible-upgrade_2024.1-infra_lxc-rockylinux-9 https://zuul.opendev.org/t/openstack/build/19583589b47c43c7bca4e6f5a7b5744d : FAILURE in 33m 28s (non-voting)\n- openstack-ansible-upgrade-infra_lxc-ubuntu-jammy https://zuul.opendev.org/t/openstack/build/13dc11406ba945a8842dff1c838d6433 : SUCCESS in 1h 35m 09s\n- openstack-ansible-upgrade_2024.1-infra_lxc-ubuntu-jammy https://zuul.opendev.org/t/openstack/build/e4224d9c904b4e27bc23b650fbeae126 : FAILURE in 27m 44s\n- openstack-ansible-deploy-hosts_lxc_zfs-ubuntu-jammy https://zuul.opendev.org/t/openstack/build/63a12cf9317f4cf88f538984875736eb : SUCCESS in 19m 52s\n- openstack-ansible-deploy-hosts_lxc_btrfs-ubuntu-jammy https://zuul.opendev.org/t/openstack/build/23cf9f40f7e44a5882c551438aea83bd : SUCCESS in 26m 11s\n- openstack-ansible-deploy-hosts_lxc_overlayfs-ubuntu-jammy https://zuul.opendev.org/t/openstack/build/1277c58943cc42a4b46f5eedfb31c897 : SUCCESS in 30m 20s\n- openstack-ansible-deploy-hosts_lxc_zfs-ubuntu-noble https://zuul.opendev.org/t/openstack/build/6adda1c9c74f4ff9b373b0d9b6138a91 : SUCCESS in 26m 41s\n- openstack-ansible-deploy-hosts_lxc_btrfs-ubuntu-noble https://zuul.opendev.org/t/openstack/build/f27128befd894e2db272a01cb041dbf3 : SUCCESS in 23m 21s\n- openstack-ansible-deploy-hosts_lxc_overlayfs-ubuntu-noble https://zuul.opendev.org/t/openstack/build/266d369e4acf4805a65404cfbd281f95 : SUCCESS in 15m 10s\n- openstack-ansible-deploy-hosts_lxc_btrfs-debian-bookworm https://zuul.opendev.org/t/openstack/build/ab7a5fee530347628da115adcbbfb360 : SUCCESS in 26m 36s\n- openstack-ansible-deploy-hosts_lxc_overlayfs-debian-bookworm https://zuul.opendev.org/t/openstack/build/86c3387fa7ec4aa8b0d06457a070ade0 : SUCCESS in 24m 42s\n- openstack-ansible-deploy-hosts_lxc_overlayfs-rockylinux-9 https://zuul.opendev.org/t/openstack/build/b9e885f02a31460382abb27ef2149a0d : SUCCESS in 27m 44s\n- openstack-ansible-deploy-hosts_lxc_overlayfs-rockylinux-10 https://zuul.opendev.org/t/openstack/build/da402dc949ee455ebc2cfdc70b777f00 : SUCCESS in 21m 23s","accounts_in_message":[],"_revision_number":3},{"id":"f07d5f65f63dcf1569e42bb95cebd6f931402143","tag":"autogenerated:gerrit:newPatchSet","author":{"_account_id":37598,"name":"Ivan Anfimov","display_name":"Ivan Anfimov","email":"lazekteam@gmail.com","username":"anfimovir"},"date":"2026-01-09 17:39:58.000000000","message":"Patch Set 4: Patch Set 3 was rebased\n\nOutdated Votes:\n* Verified-1\n","accounts_in_message":[],"_revision_number":4},{"id":"68736e2d1bf8973c0dddd6cc3fc012a0b85af7a6","tag":"autogenerated:zuul:check","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2026-01-09 19:27:15.000000000","message":"Patch Set 4: Verified-1\n\nBuild failed (check pipeline).  For information on how to proceed, see\nhttps://docs.opendev.org/opendev/infra-manual/latest/developers.html#automated-testing\nand https://docs.openstack.org/project-team-guide/testing.html#how-to-handle-test-failures\n\nhttps://zuul.opendev.org/t/openstack/buildset/4234b597f63442128cd91e9e4981d26f\n\n- openstack-ansible-linters-ubuntu-noble https://zuul.opendev.org/t/openstack/build/ada55f201e67437db6a01f49319a8738 : SUCCESS in 7m 08s\n- openstack-tox-docs https://zuul.opendev.org/t/openstack/build/f294e4def3174597b040097e13ef491d : SUCCESS in 5m 08s\n- openstack-ansible-deploy-infra_lxc-centos-9-stream https://zuul.opendev.org/t/openstack/build/50a4ccce6a6c4494a5fd181e9b008e7c : FAILURE in 38m 35s (non-voting)\n- openstack-ansible-deploy-infra_lxc-rockylinux-9 https://zuul.opendev.org/t/openstack/build/9dee06ea68194d3dab4639fb2f231e0b : FAILURE in 30m 28s (non-voting)\n- openstack-ansible-deploy-infra_lxc_tls-rockylinux-9 https://zuul.opendev.org/t/openstack/build/927a9063e5034208b545c12bbcbb2dd1 : FAILURE in 51m 54s (non-voting)\n- openstack-ansible-deploy-infra_lxc-rockylinux-10 https://zuul.opendev.org/t/openstack/build/b100d3f9aff24609a1dbb3a4558f70d9 : FAILURE in 53m 39s (non-voting)\n- openstack-ansible-deploy-infra_lxc_tls-rockylinux-10 https://zuul.opendev.org/t/openstack/build/84c97f9e0a6244b0988c9c89621f434f : FAILURE in 51m 20s (non-voting)\n- openstack-ansible-deploy-infra_lxc-centos-10-stream https://zuul.opendev.org/t/openstack/build/a092af94016a4b58bf1f0e85bfa4cec5 : FAILURE in 48m 26s (non-voting)\n- openstack-ansible-deploy-infra_lxc-debian-bookworm https://zuul.opendev.org/t/openstack/build/7326db230a4842ad9533038a9438796c : SUCCESS in 1h 12m 08s\n- openstack-ansible-deploy-infra_lxc-ubuntu-jammy https://zuul.opendev.org/t/openstack/build/6e7c7160b4ba4356b2c2bd8e2dc100ca : SUCCESS in 1h 05m 53s\n- openstack-ansible-deploy-infra_lxc_stepca-ubuntu-jammy https://zuul.opendev.org/t/openstack/build/3d846ef1a5dc4f178e2b8599fcc813a7 : SUCCESS in 1h 14m 48s\n- openstack-ansible-deploy-infra_lxc_tls-ubuntu-jammy https://zuul.opendev.org/t/openstack/build/b3540b3149454aff946f0e87e539ad2c : SUCCESS in 1h 09m 32s\n- openstack-ansible-deploy-infra_lxc-ubuntu-noble https://zuul.opendev.org/t/openstack/build/e9678917d4064785b2f6d1fe99680b12 : SUCCESS in 1h 22m 36s\n- openstack-ansible-deploy-infra_lxc_stepca-ubuntu-noble https://zuul.opendev.org/t/openstack/build/986148bd7cc644758eed528fe3169b39 : SUCCESS in 1h 23m 19s\n- openstack-ansible-deploy-infra_lxc_tls-ubuntu-noble https://zuul.opendev.org/t/openstack/build/047c3257c8be4866977bf2de5c43d780 : SUCCESS in 1h 25m 50s\n- openstack-ansible-upgrade-infra_lxc-rockylinux-9 https://zuul.opendev.org/t/openstack/build/4d679870434f416397950f5d18e49d6e : FAILURE in 44m 05s (non-voting)\n- openstack-ansible-upgrade_2024.1-infra_lxc-rockylinux-9 https://zuul.opendev.org/t/openstack/build/456ce69246724c2197f5ec2c33dabfdb : FAILURE in 33m 57s (non-voting)\n- openstack-ansible-upgrade-infra_lxc-ubuntu-jammy https://zuul.opendev.org/t/openstack/build/a0353163b8054325bad7e8813679dc3c : SUCCESS in 1h 38m 45s\n- openstack-ansible-upgrade_2024.1-infra_lxc-ubuntu-jammy https://zuul.opendev.org/t/openstack/build/51c339c2ca0c45c19974dcad2172f098 : FAILURE in 24m 59s\n- openstack-ansible-deploy-hosts_lxc_zfs-ubuntu-jammy https://zuul.opendev.org/t/openstack/build/0ff0da53def94e989d151323aee00415 : SUCCESS in 26m 11s\n- openstack-ansible-deploy-hosts_lxc_btrfs-ubuntu-jammy https://zuul.opendev.org/t/openstack/build/ee6c3734e5534386819adf739f7a70e3 : SUCCESS in 24m 19s\n- openstack-ansible-deploy-hosts_lxc_overlayfs-ubuntu-jammy https://zuul.opendev.org/t/openstack/build/03d7e1a939464b0b98ea4f9f18393af5 : SUCCESS in 28m 12s\n- openstack-ansible-deploy-hosts_lxc_zfs-ubuntu-noble https://zuul.opendev.org/t/openstack/build/5626fc4fd0a34e45a40770c4a11a5111 : SUCCESS in 23m 42s\n- openstack-ansible-deploy-hosts_lxc_btrfs-ubuntu-noble https://zuul.opendev.org/t/openstack/build/96523d2d98ef4f0b924727a3b5a9174c : SUCCESS in 27m 00s\n- openstack-ansible-deploy-hosts_lxc_overlayfs-ubuntu-noble https://zuul.opendev.org/t/openstack/build/b40ad30a609a41a49fd1bad175fa3c54 : SUCCESS in 26m 10s\n- openstack-ansible-deploy-hosts_lxc_btrfs-debian-bookworm https://zuul.opendev.org/t/openstack/build/a46dd55ddf1d4240b4f3d95fac417c93 : SUCCESS in 26m 22s\n- openstack-ansible-deploy-hosts_lxc_overlayfs-debian-bookworm https://zuul.opendev.org/t/openstack/build/3a2f57c549c44758952119a89cea6228 : SUCCESS in 26m 23s\n- openstack-ansible-deploy-hosts_lxc_overlayfs-rockylinux-9 https://zuul.opendev.org/t/openstack/build/67cf92509dff45acacd73a3deed4a390 : SUCCESS in 26m 54s\n- openstack-ansible-deploy-hosts_lxc_overlayfs-rockylinux-10 https://zuul.opendev.org/t/openstack/build/3b46ec64f376411b94076602006a3bd9 : SUCCESS in 18m 32s","accounts_in_message":[],"_revision_number":4},{"id":"a7004a99a8ffc283655831b4d765c612cd5f5563","author":{"_account_id":37598,"name":"Ivan Anfimov","display_name":"Ivan Anfimov","email":"lazekteam@gmail.com","username":"anfimovir"},"date":"2026-01-09 19:37:41.000000000","message":"Patch Set 4:\n\n(1 comment)","accounts_in_message":[],"_revision_number":4},{"id":"118e7898a13acd5295ecf1a6c4ea0f3fee40f2c6","tag":"autogenerated:gerrit:newPatchSet","author":{"_account_id":37598,"name":"Ivan Anfimov","display_name":"Ivan Anfimov","email":"lazekteam@gmail.com","username":"anfimovir"},"date":"2026-01-09 19:44:06.000000000","message":"Patch Set 5: Commit message was updated.\n\nOutdated Votes:\n* Verified-1\n","accounts_in_message":[],"_revision_number":5},{"id":"a33177e78a524b053f2abb5640582f0dae1472d9","author":{"_account_id":37598,"name":"Ivan Anfimov","display_name":"Ivan Anfimov","email":"lazekteam@gmail.com","username":"anfimovir"},"date":"2026-01-09 19:44:39.000000000","message":"Patch Set 5:\n\n(1 comment)","accounts_in_message":[],"_revision_number":5},{"id":"ae8d05bd3afdfc498f9a2114316bea8eae0eb2a1","tag":"autogenerated:gerrit:newPatchSet","author":{"_account_id":37598,"name":"Ivan Anfimov","display_name":"Ivan Anfimov","email":"lazekteam@gmail.com","username":"anfimovir"},"date":"2026-01-09 19:45:55.000000000","message":"Patch Set 6: Patch Set 5 was rebased","accounts_in_message":[],"_revision_number":6},{"id":"d74cd53ebf2b1ce0670017afebb10b87a7c52a91","tag":"autogenerated:zuul:check","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2026-01-09 21:53:23.000000000","message":"Patch Set 6: Verified+1\n\nBuild succeeded (check pipeline).\nhttps://zuul.opendev.org/t/openstack/buildset/de922b98965140fa9bb49634e225a2f9\n\n- openstack-ansible-linters-ubuntu-noble https://zuul.opendev.org/t/openstack/build/33e49720564e475a9a7b17b96bf46b8d : SUCCESS in 10m 23s\n- openstack-tox-docs https://zuul.opendev.org/t/openstack/build/4d3851923d384a94bb33464e62fce1db : SUCCESS in 6m 25s\n- openstack-ansible-deploy-infra_lxc-centos-9-stream https://zuul.opendev.org/t/openstack/build/ae80a833718b459dba7fc4bf6f23ed39 : FAILURE in 29m 00s (non-voting)\n- openstack-ansible-deploy-infra_lxc-rockylinux-9 https://zuul.opendev.org/t/openstack/build/c9bd4e26b2544e27a7b7c8f713f43181 : FAILURE in 45m 06s (non-voting)\n- openstack-ansible-deploy-infra_lxc_tls-rockylinux-9 https://zuul.opendev.org/t/openstack/build/219f31aad83d44af9c82838527f99481 : FAILURE in 28m 44s (non-voting)\n- openstack-ansible-deploy-infra_lxc-rockylinux-10 https://zuul.opendev.org/t/openstack/build/e4755e80b46f4ff5a2f051c256179f3b : FAILURE in 30m 25s (non-voting)\n- openstack-ansible-deploy-infra_lxc_tls-rockylinux-10 https://zuul.opendev.org/t/openstack/build/6cd7d6fe8ed24bdaa070eeab87ac9668 : FAILURE in 29m 18s (non-voting)\n- openstack-ansible-deploy-infra_lxc-centos-10-stream https://zuul.opendev.org/t/openstack/build/87f7d783efce4f73a85836760346e2aa : FAILURE in 40m 15s (non-voting)\n- openstack-ansible-deploy-infra_lxc-debian-bookworm https://zuul.opendev.org/t/openstack/build/55c017121c1547698a6e5da28ea52e75 : SUCCESS in 1h 11m 47s\n- openstack-ansible-deploy-infra_lxc-ubuntu-jammy https://zuul.opendev.org/t/openstack/build/0228609859e74d3281529c9fed51b797 : SUCCESS in 1h 08m 33s\n- openstack-ansible-deploy-infra_lxc_stepca-ubuntu-jammy https://zuul.opendev.org/t/openstack/build/6105d2e691484f1c8006fe4339b74b99 : SUCCESS in 1h 11m 15s\n- openstack-ansible-deploy-infra_lxc_tls-ubuntu-jammy https://zuul.opendev.org/t/openstack/build/b0dcd951a7c14feab0ff952733adde67 : SUCCESS in 1h 12m 55s\n- openstack-ansible-deploy-infra_lxc-ubuntu-noble https://zuul.opendev.org/t/openstack/build/a678a608f9584bffb20bdf8ee4d56692 : SUCCESS in 1h 22m 49s\n- openstack-ansible-deploy-infra_lxc_stepca-ubuntu-noble https://zuul.opendev.org/t/openstack/build/5718c9d6bbcd45028b0af3f63d7e45a6 : SUCCESS in 1h 12m 54s\n- openstack-ansible-deploy-infra_lxc_tls-ubuntu-noble https://zuul.opendev.org/t/openstack/build/6007cb99ae1b485eb8c31a81b190a390 : SUCCESS in 44m 25s\n- openstack-ansible-upgrade-infra_lxc-rockylinux-9 https://zuul.opendev.org/t/openstack/build/058565deb2114050bc547e5d952c2e14 : FAILURE in 46m 15s (non-voting)\n- openstack-ansible-upgrade_2024.1-infra_lxc-rockylinux-9 https://zuul.opendev.org/t/openstack/build/baa1a23fd00a4e418ce2079480503a8b : FAILURE in 43m 17s (non-voting)\n- openstack-ansible-upgrade-infra_lxc-ubuntu-jammy https://zuul.opendev.org/t/openstack/build/92d0403f0fd1457cba3d0cef31d98478 : SUCCESS in 1h 34m 29s\n- openstack-ansible-upgrade_2024.1-infra_lxc-ubuntu-jammy https://zuul.opendev.org/t/openstack/build/c77e1175e2b74fc1a6a2542184dfae65 : SUCCESS in 1h 39m 44s\n- openstack-ansible-deploy-hosts_lxc_zfs-ubuntu-jammy https://zuul.opendev.org/t/openstack/build/93fca72b1f78459da4652124cb3a1b31 : SUCCESS in 15m 02s\n- openstack-ansible-deploy-hosts_lxc_btrfs-ubuntu-jammy https://zuul.opendev.org/t/openstack/build/562302ab2cce40718a74607b87400d9a : SUCCESS in 15m 21s\n- openstack-ansible-deploy-hosts_lxc_overlayfs-ubuntu-jammy https://zuul.opendev.org/t/openstack/build/d411d19d8ba34fc49c5b703edf6f59cc : SUCCESS in 25m 27s\n- openstack-ansible-deploy-hosts_lxc_zfs-ubuntu-noble https://zuul.opendev.org/t/openstack/build/6ed6d684355d497cb6c0b5ce974bb352 : SUCCESS in 23m 18s\n- openstack-ansible-deploy-hosts_lxc_btrfs-ubuntu-noble https://zuul.opendev.org/t/openstack/build/eef584c1267f4ff8ae6433a378c60ca9 : SUCCESS in 23m 46s\n- openstack-ansible-deploy-hosts_lxc_overlayfs-ubuntu-noble https://zuul.opendev.org/t/openstack/build/94e0ea435a9e4937bb71a5a47b123465 : SUCCESS in 23m 18s\n- openstack-ansible-deploy-hosts_lxc_btrfs-debian-bookworm https://zuul.opendev.org/t/openstack/build/74562cb4f0014027a647a43edccde0fe : SUCCESS in 26m 07s\n- openstack-ansible-deploy-hosts_lxc_overlayfs-debian-bookworm https://zuul.opendev.org/t/openstack/build/4e2c56a410c245c8a172f05f2b141ba1 : SUCCESS in 26m 08s\n- openstack-ansible-deploy-hosts_lxc_overlayfs-rockylinux-9 https://zuul.opendev.org/t/openstack/build/87d7b1e655324eb8a36ee7b7ffafef2a : SUCCESS in 26m 55s\n- openstack-ansible-deploy-hosts_lxc_overlayfs-rockylinux-10 https://zuul.opendev.org/t/openstack/build/1a016c4630584cc2adb26bd7c7003311 : SUCCESS in 25m 10s","accounts_in_message":[],"_revision_number":6},{"id":"7756688bd7a879e68d027888d8a9860b9d610535","author":{"_account_id":31542,"name":"Andrew Bonney","email":"andrew.bonney@bbc.co.uk","username":"andrewbonney"},"date":"2026-01-12 08:19:02.000000000","message":"Patch Set 6: Code-Review+2","accounts_in_message":[],"_revision_number":6},{"id":"f3c14db462d781e002401bc41fdc5afe8d5ef7f1","tag":"autogenerated:gerrit:newPatchSet","author":{"_account_id":28619,"name":"Dmitriy Rabotyagov","email":"noonedeadpunk@gmail.com","username":"noonedeadpunk"},"date":"2026-01-16 15:10:46.000000000","message":"Patch Set 7: Cherry Picked from branch master.\n\nOutdated Votes:\n* Code-Review+2 (copy condition: \"changekind:TRIVIAL_REBASE OR is:MIN\")\n* Verified+1\n","accounts_in_message":[],"_revision_number":7},{"id":"ae5514f27705af4b14fe07d41f8952d2f817fb38","tag":"autogenerated:gerrit:newPatchSet","author":{"_account_id":28619,"name":"Dmitriy Rabotyagov","email":"noonedeadpunk@gmail.com","username":"noonedeadpunk"},"date":"2026-01-16 15:10:59.000000000","message":"Patch Set 8: Patch Set 7 was rebased","accounts_in_message":[],"_revision_number":8},{"id":"18741da52f91e4398dabbd63c98c9def9509d0b7","tag":"autogenerated:gerrit:newPatchSet","author":{"_account_id":28619,"name":"Dmitriy Rabotyagov","email":"noonedeadpunk@gmail.com","username":"noonedeadpunk"},"date":"2026-01-16 17:08:22.000000000","message":"Patch Set 9: Patch Set 8 was rebased","accounts_in_message":[],"_revision_number":9},{"id":"e615b0533e44c90313263bd0c654e9a8bcf44267","tag":"autogenerated:zuul:check","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2026-01-16 19:20:41.000000000","message":"Patch Set 9: Verified-1\n\nBuild failed (check pipeline).  For information on how to proceed, see\nhttps://docs.opendev.org/opendev/infra-manual/latest/developers.html#automated-testing\nand https://docs.openstack.org/project-team-guide/testing.html#how-to-handle-test-failures\n\nhttps://zuul.opendev.org/t/openstack/buildset/1ccc776cf93a426d92ccaf7ecc4036c6\n\n- openstack-ansible-linters-ubuntu-noble https://zuul.opendev.org/t/openstack/build/16dddbfec82b40419094cbc067bb2a04 : SUCCESS in 14m 33s\n- openstack-tox-docs https://zuul.opendev.org/t/openstack/build/caa923481fb24d84b03fbb235dce767e : SUCCESS in 6m 07s\n- openstack-ansible-deploy-infra_lxc-centos-9-stream https://zuul.opendev.org/t/openstack/build/3d39a83f5c824742845af63775730531 : FAILURE in 50m 12s (non-voting)\n- openstack-ansible-deploy-infra_lxc-rockylinux-9 https://zuul.opendev.org/t/openstack/build/3887e1eeb67f46caadd1f0a702ef276f : FAILURE in 49m 11s (non-voting)\n- openstack-ansible-deploy-infra_lxc_tls-rockylinux-9 https://zuul.opendev.org/t/openstack/build/1690694353554f5586b31275c69c0565 : FAILURE in 50m 48s (non-voting)\n- openstack-ansible-deploy-infra_lxc-rockylinux-10 https://zuul.opendev.org/t/openstack/build/5788ce9367b44a6fb0d8f44f849d238d : FAILURE in 46m 48s (non-voting)\n- openstack-ansible-deploy-infra_lxc_tls-rockylinux-10 https://zuul.opendev.org/t/openstack/build/0cc2bf36d6654ff4b624908e0a6c52e9 : FAILURE in 34m 10s (non-voting)\n- openstack-ansible-deploy-infra_lxc-centos-10-stream https://zuul.opendev.org/t/openstack/build/1b2cb77166cc4f8f9b10149a69aed78b : FAILURE in 43m 03s (non-voting)\n- openstack-ansible-deploy-infra_lxc-debian-bookworm https://zuul.opendev.org/t/openstack/build/557acda62bc64fad8dd6a1e4d5f49ad0 : SUCCESS in 1h 15m 18s\n- openstack-ansible-deploy-infra_lxc-ubuntu-jammy https://zuul.opendev.org/t/openstack/build/aa67350426a24db799af361bdf42496a : SUCCESS in 1h 20m 13s\n- openstack-ansible-deploy-infra_lxc_stepca-ubuntu-jammy https://zuul.opendev.org/t/openstack/build/6492ad2da4e84043906f4fca87a8876d : SUCCESS in 1h 33m 34s\n- openstack-ansible-deploy-infra_lxc_tls-ubuntu-jammy https://zuul.opendev.org/t/openstack/build/3aaa2a0075874afdbaaa79bd8ed324eb : SUCCESS in 1h 17m 42s\n- openstack-ansible-deploy-infra_lxc-ubuntu-noble https://zuul.opendev.org/t/openstack/build/a265909b60844598866fe396cfc8eff1 : SUCCESS in 46m 04s\n- openstack-ansible-deploy-infra_lxc_stepca-ubuntu-noble https://zuul.opendev.org/t/openstack/build/68a44a2f11fb45f09049bafa365a8ccd : SUCCESS in 50m 26s\n- openstack-ansible-deploy-infra_lxc_tls-ubuntu-noble https://zuul.opendev.org/t/openstack/build/dd5db6e21b3d49a0ab7521ced0e1c856 : SUCCESS in 1h 14m 27s\n- openstack-ansible-upgrade-infra_lxc-rockylinux-9 https://zuul.opendev.org/t/openstack/build/c592ed11b148418e97da9cf530a64f9a : SUCCESS in 1h 04m 39s\n- openstack-ansible-upgrade_2024.1-infra_lxc-rockylinux-9 https://zuul.opendev.org/t/openstack/build/7f1008856d494743ac1a339c4227d612 : FAILURE in 28m 07s\n- openstack-ansible-upgrade-infra_lxc-ubuntu-jammy https://zuul.opendev.org/t/openstack/build/57454a6cd6e94a62a7c54772d9468773 : SUCCESS in 1h 30m 56s\n- openstack-ansible-upgrade_2024.1-infra_lxc-ubuntu-jammy https://zuul.opendev.org/t/openstack/build/08e5412548e74be1878029081ea8f7e4 : SUCCESS in 2h 07m 33s\n- openstack-ansible-deploy-hosts_lxc_zfs-ubuntu-jammy https://zuul.opendev.org/t/openstack/build/ad5275c25c9e4733b5e06821a7bab088 : SUCCESS in 21m 41s\n- openstack-ansible-deploy-hosts_lxc_btrfs-ubuntu-jammy https://zuul.opendev.org/t/openstack/build/335b17c77bac40309586c317f780345c : SUCCESS in 26m 13s\n- openstack-ansible-deploy-hosts_lxc_overlayfs-ubuntu-jammy https://zuul.opendev.org/t/openstack/build/7f53285f914b4aee8b8886145a0c0c63 : SUCCESS in 20m 50s\n- openstack-ansible-deploy-hosts_lxc_zfs-ubuntu-noble https://zuul.opendev.org/t/openstack/build/bc581de9bc8a466582babbbac390c40c : SUCCESS in 26m 03s\n- openstack-ansible-deploy-hosts_lxc_btrfs-ubuntu-noble https://zuul.opendev.org/t/openstack/build/a631d55a93d74b019b48576995342c70 : SUCCESS in 31m 01s\n- openstack-ansible-deploy-hosts_lxc_overlayfs-ubuntu-noble https://zuul.opendev.org/t/openstack/build/1c431befeb8845e0bcd02fa0a4088e7a : SUCCESS in 47m 04s\n- openstack-ansible-deploy-hosts_lxc_btrfs-debian-bookworm https://zuul.opendev.org/t/openstack/build/a422868cba3d41d0b3f8aa01408c002d : SUCCESS in 27m 50s\n- openstack-ansible-deploy-hosts_lxc_overlayfs-debian-bookworm https://zuul.opendev.org/t/openstack/build/8fe485c876b54f4cb2ca3e4059ab242b : SUCCESS in 25m 28s\n- openstack-ansible-deploy-hosts_lxc_overlayfs-rockylinux-9 https://zuul.opendev.org/t/openstack/build/5d608bfa0c404521904102649fb18b48 : SUCCESS in 33m 04s\n- openstack-ansible-deploy-hosts_lxc_overlayfs-rockylinux-10 https://zuul.opendev.org/t/openstack/build/a4dd245055524aedb445d01588b5e6e0 : SUCCESS in 35m 28s","accounts_in_message":[],"_revision_number":9},{"id":"3aa1b303f1f4456ab6e6ca7d66fe999a52d700a9","author":{"_account_id":28619,"name":"Dmitriy Rabotyagov","email":"noonedeadpunk@gmail.com","username":"noonedeadpunk"},"date":"2026-01-19 08:15:23.000000000","message":"Patch Set 9:\n\n(1 comment)","accounts_in_message":[],"_revision_number":9},{"id":"4eb71153aa319d74aa058310abbbd2be9e914ae5","tag":"autogenerated:zuul:check","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2026-01-19 10:30:24.000000000","message":"Patch Set 9: Verified+1\n\nBuild succeeded (check pipeline).\nhttps://zuul.opendev.org/t/openstack/buildset/6871c3fb48e646e3ac39482b7ed0ebb3\n\n- openstack-ansible-linters-ubuntu-noble https://zuul.opendev.org/t/openstack/build/8e09c56afd1149a9a3de61ecb1bb0cd9 : SUCCESS in 12m 42s\n- openstack-tox-docs https://zuul.opendev.org/t/openstack/build/e9fcae08aa8a4c1698b4b5a8030cf02a : SUCCESS in 5m 59s\n- openstack-ansible-deploy-infra_lxc-centos-9-stream https://zuul.opendev.org/t/openstack/build/55e60c85ae944b0ab42e3c85e9152de3 : FAILURE in 49m 00s (non-voting)\n- openstack-ansible-deploy-infra_lxc-rockylinux-9 https://zuul.opendev.org/t/openstack/build/70a41f8ab25440be960e35351c2b8820 : FAILURE in 57m 43s (non-voting)\n- openstack-ansible-deploy-infra_lxc_tls-rockylinux-9 https://zuul.opendev.org/t/openstack/build/09dd79ee348245f388e31ef6905bbb1a : FAILURE in 50m 59s (non-voting)\n- openstack-ansible-deploy-infra_lxc-rockylinux-10 https://zuul.opendev.org/t/openstack/build/a74ea71c194543b4b105ddd4877b56c9 : FAILURE in 33m 18s (non-voting)\n- openstack-ansible-deploy-infra_lxc_tls-rockylinux-10 https://zuul.opendev.org/t/openstack/build/149badedbd064cedb1682ff8796d00bc : FAILURE in 41m 23s (non-voting)\n- openstack-ansible-deploy-infra_lxc-centos-10-stream https://zuul.opendev.org/t/openstack/build/18e39b6e435e47bf80ebfb79f4276af8 : FAILURE in 47m 44s (non-voting)\n- openstack-ansible-deploy-infra_lxc-debian-bookworm https://zuul.opendev.org/t/openstack/build/33e70df7fab34435864a6374a6fbdffc : SUCCESS in 1h 21m 33s\n- openstack-ansible-deploy-infra_lxc-ubuntu-jammy https://zuul.opendev.org/t/openstack/build/4f777f3892cf401fa0619a3774ea0984 : SUCCESS in 1h 14m 41s\n- openstack-ansible-deploy-infra_lxc_stepca-ubuntu-jammy https://zuul.opendev.org/t/openstack/build/c6c4faa2c61f4da6b2e1cf7154e5f940 : SUCCESS in 1h 21m 13s\n- openstack-ansible-deploy-infra_lxc_tls-ubuntu-jammy https://zuul.opendev.org/t/openstack/build/9be8ded7cd744fc891fb2df5115447e0 : SUCCESS in 1h 28m 05s\n- openstack-ansible-deploy-infra_lxc-ubuntu-noble https://zuul.opendev.org/t/openstack/build/d6ffdf454b59494b9054c569a628136e : SUCCESS in 1h 03m 58s\n- openstack-ansible-deploy-infra_lxc_stepca-ubuntu-noble https://zuul.opendev.org/t/openstack/build/ed19015ab214430394750958f5db82b2 : SUCCESS in 1h 18m 13s\n- openstack-ansible-deploy-infra_lxc_tls-ubuntu-noble https://zuul.opendev.org/t/openstack/build/36f32323297c4f19807b83029a18d0a4 : SUCCESS in 1h 15m 05s\n- openstack-ansible-upgrade-infra_lxc-rockylinux-9 https://zuul.opendev.org/t/openstack/build/1f18b4539e6e4908a89a530b638189d8 : SUCCESS in 1h 55m 17s\n- openstack-ansible-upgrade_2024.1-infra_lxc-rockylinux-9 https://zuul.opendev.org/t/openstack/build/01dce5a63fdd4db3a00a03f8df657ca5 : SUCCESS in 2h 05m 17s\n- openstack-ansible-upgrade-infra_lxc-ubuntu-jammy https://zuul.opendev.org/t/openstack/build/136b162476bf4769ba825997635e45bb : SUCCESS in 51m 03s\n- openstack-ansible-upgrade_2024.1-infra_lxc-ubuntu-jammy https://zuul.opendev.org/t/openstack/build/3d45cd40efbd4f929e5fc5f171a73066 : SUCCESS in 56m 22s\n- openstack-ansible-deploy-hosts_lxc_zfs-ubuntu-jammy https://zuul.opendev.org/t/openstack/build/7bbcbdf1ef6e4a9a9db2e9f918485286 : SUCCESS in 26m 25s\n- openstack-ansible-deploy-hosts_lxc_btrfs-ubuntu-jammy https://zuul.opendev.org/t/openstack/build/ab133dbd237840d29dd674604ffe2866 : SUCCESS in 26m 23s\n- openstack-ansible-deploy-hosts_lxc_overlayfs-ubuntu-jammy https://zuul.opendev.org/t/openstack/build/2a8ae84cc83e42c5b609460ae0fb38ba : SUCCESS in 31m 34s\n- openstack-ansible-deploy-hosts_lxc_zfs-ubuntu-noble https://zuul.opendev.org/t/openstack/build/b7f2e7e31ce9425380c026d6f645e92a : SUCCESS in 17m 12s\n- openstack-ansible-deploy-hosts_lxc_btrfs-ubuntu-noble https://zuul.opendev.org/t/openstack/build/23867d2343d849a48eadc8aeaa15552e : SUCCESS in 27m 31s\n- openstack-ansible-deploy-hosts_lxc_overlayfs-ubuntu-noble https://zuul.opendev.org/t/openstack/build/6b07fce4c64b43cda0bcdace8c8a4c65 : SUCCESS in 27m 14s\n- openstack-ansible-deploy-hosts_lxc_btrfs-debian-bookworm https://zuul.opendev.org/t/openstack/build/34e61be8f14d473da924e153f178214a : SUCCESS in 28m 43s\n- openstack-ansible-deploy-hosts_lxc_overlayfs-debian-bookworm https://zuul.opendev.org/t/openstack/build/b053e141cc7c47a0ae2b45822b2fcd06 : SUCCESS in 28m 13s\n- openstack-ansible-deploy-hosts_lxc_overlayfs-rockylinux-9 https://zuul.opendev.org/t/openstack/build/7e7a3f23a41b4dc1b01e018fa88cfeb5 : SUCCESS in 30m 38s\n- openstack-ansible-deploy-hosts_lxc_overlayfs-rockylinux-10 https://zuul.opendev.org/t/openstack/build/673d22b3f18a49a7acd7ae5259f0104a : SUCCESS in 17m 45s","accounts_in_message":[],"_revision_number":9},{"id":"c5f1dae53c3b3e32683a3a9dd8e5ea3e5bda4381","author":{"_account_id":37598,"name":"Ivan Anfimov","display_name":"Ivan Anfimov","email":"lazekteam@gmail.com","username":"anfimovir"},"date":"2026-01-19 10:30:59.000000000","message":"Patch Set 9: Code-Review+1","accounts_in_message":[],"_revision_number":9},{"id":"ffa83198ce80b257783bac100b8f87260f4bd69d","author":{"_account_id":31542,"name":"Andrew Bonney","email":"andrew.bonney@bbc.co.uk","username":"andrewbonney"},"date":"2026-01-19 11:02:55.000000000","message":"Patch Set 9: Code-Review+2","accounts_in_message":[],"_revision_number":9},{"id":"c2062faa40774ff8a93235457b64c05577044592","author":{"_account_id":28619,"name":"Dmitriy Rabotyagov","email":"noonedeadpunk@gmail.com","username":"noonedeadpunk"},"date":"2026-01-19 11:10:41.000000000","message":"Patch Set 9: Code-Review+2 Workflow+1","accounts_in_message":[],"_revision_number":9},{"id":"e81d2aabbddb370743349f7f345d792f44db57a5","tag":"autogenerated:zuul:gate","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2026-01-19 11:13:21.000000000","message":"Patch Set 9: -Verified\n\nStarting gate jobs.","accounts_in_message":[],"_revision_number":9},{"id":"e6ce9298b25910168c18d5d5bf1687a86baa25de","tag":"autogenerated:zuul:gate","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2026-01-19 13:31:07.000000000","message":"Patch Set 9: Verified+2\n\nBuild succeeded (gate pipeline).\nhttps://zuul.opendev.org/t/openstack/buildset/a80a8d18b9264900898729f26ede13e5\n\n- openstack-ansible-linters-ubuntu-noble https://zuul.opendev.org/t/openstack/build/7db0f2bb519d42a2a7e45390c2a79420 : SUCCESS in 12m 02s\n- openstack-tox-docs https://zuul.opendev.org/t/openstack/build/c0c56c760dc04a06b42c4268f522b8eb : SUCCESS in 6m 15s\n- openstack-ansible-deploy-infra_lxc-rockylinux-9 https://zuul.opendev.org/t/openstack/build/a0cbf75125c54eea9d5703aa3bb5da0c : FAILURE in 43m 49s (non-voting)\n- openstack-ansible-deploy-infra_lxc_tls-rockylinux-9 https://zuul.opendev.org/t/openstack/build/51a57e7f52ad484c9089b219d564e962 : FAILURE in 44m 17s (non-voting)\n- openstack-ansible-deploy-infra_lxc-rockylinux-10 https://zuul.opendev.org/t/openstack/build/af18dcf72e9b41d3b999700353edbb49 : FAILURE in 31m 59s (non-voting)\n- openstack-ansible-deploy-infra_lxc_tls-rockylinux-10 https://zuul.opendev.org/t/openstack/build/a5b4224b790044e4bf16efb487921454 : FAILURE in 33m 40s (non-voting)\n- openstack-ansible-deploy-infra_lxc-centos-10-stream https://zuul.opendev.org/t/openstack/build/9fa6b7691ec748089cd5c35109a834d9 : FAILURE in 51m 23s (non-voting)\n- openstack-ansible-deploy-infra_lxc-debian-bookworm https://zuul.opendev.org/t/openstack/build/7c66daeb534a4b73b29f388b6488b478 : SUCCESS in 1h 16m 48s\n- openstack-ansible-deploy-infra_lxc-ubuntu-jammy https://zuul.opendev.org/t/openstack/build/8646973f35a149baab79144199594c72 : SUCCESS in 48m 06s\n- openstack-ansible-deploy-infra_lxc_stepca-ubuntu-jammy https://zuul.opendev.org/t/openstack/build/960cf95b4fea4529bc397b5b1fda6f28 : SUCCESS in 1h 12m 07s\n- openstack-ansible-deploy-infra_lxc_tls-ubuntu-jammy https://zuul.opendev.org/t/openstack/build/e5ae9bbe973b4cae9e0ff565238e72e0 : SUCCESS in 1h 06m 49s\n- openstack-ansible-deploy-infra_lxc-ubuntu-noble https://zuul.opendev.org/t/openstack/build/514ce280213b41f7b9cba28d72f3a446 : SUCCESS in 1h 18m 27s\n- openstack-ansible-deploy-infra_lxc_stepca-ubuntu-noble https://zuul.opendev.org/t/openstack/build/207653b0067b4e408e93754b278f8339 : SUCCESS in 1h 08m 04s\n- openstack-ansible-deploy-infra_lxc_tls-ubuntu-noble https://zuul.opendev.org/t/openstack/build/9bba1917b3634853a4be2ec76a58982c : SUCCESS in 1h 17m 17s\n- openstack-ansible-upgrade-infra_lxc-rockylinux-9 https://zuul.opendev.org/t/openstack/build/e6b09d10bb85409faa6d4c559c8e61b9 : SUCCESS in 2h 09m 06s\n- openstack-ansible-upgrade_2024.1-infra_lxc-rockylinux-9 https://zuul.opendev.org/t/openstack/build/df29b306927d44d192893580d8353a4c : SUCCESS in 1h 10m 31s\n- openstack-ansible-upgrade-infra_lxc-ubuntu-jammy https://zuul.opendev.org/t/openstack/build/86bc2045ec6841d48d5806fa30cbbb83 : SUCCESS in 55m 06s\n- openstack-ansible-upgrade_2024.1-infra_lxc-ubuntu-jammy https://zuul.opendev.org/t/openstack/build/b290541969434540b4e028a81a895f96 : SUCCESS in 1h 50m 35s\n- openstack-ansible-deploy-hosts_lxc_zfs-ubuntu-jammy https://zuul.opendev.org/t/openstack/build/28166d8e86f54ba0ad0af34a57dc3ff8 : SUCCESS in 18m 10s\n- openstack-ansible-deploy-hosts_lxc_btrfs-ubuntu-jammy https://zuul.opendev.org/t/openstack/build/5d74eedf5fd94befbe815d8ac81520ac : SUCCESS in 27m 05s\n- openstack-ansible-deploy-hosts_lxc_overlayfs-ubuntu-jammy https://zuul.opendev.org/t/openstack/build/06845b26b00346a69bf8c05dee979f6a : SUCCESS in 20m 15s\n- openstack-ansible-deploy-hosts_lxc_zfs-ubuntu-noble https://zuul.opendev.org/t/openstack/build/18c294a12b0f4066847cefbd7d0bf41a : SUCCESS in 28m 01s\n- openstack-ansible-deploy-hosts_lxc_btrfs-ubuntu-noble https://zuul.opendev.org/t/openstack/build/60f96000d1274f3eac1bfa3ef4bb365d : SUCCESS in 25m 04s\n- openstack-ansible-deploy-hosts_lxc_overlayfs-ubuntu-noble https://zuul.opendev.org/t/openstack/build/e55d7a81bbcd4906a90e8d4d3a0e8dda : SUCCESS in 25m 29s\n- openstack-ansible-deploy-hosts_lxc_btrfs-debian-bookworm https://zuul.opendev.org/t/openstack/build/5003dea9b8c54af896ad6d60a7deda09 : SUCCESS in 24m 01s\n- openstack-ansible-deploy-hosts_lxc_overlayfs-debian-bookworm https://zuul.opendev.org/t/openstack/build/2aa597ef1b0d40f08416adc31d863c87 : SUCCESS in 19m 21s\n- openstack-ansible-deploy-hosts_lxc_overlayfs-rockylinux-9 https://zuul.opendev.org/t/openstack/build/e7741f913f3f4ecfbf07ae17ce7ea36c : SUCCESS in 30m 36s\n- openstack-ansible-deploy-hosts_lxc_overlayfs-rockylinux-10 https://zuul.opendev.org/t/openstack/build/1dcda3c4ece04a18a02a471240f74c95 : SUCCESS in 20m 25s","accounts_in_message":[],"_revision_number":9},{"id":"37f7ae4cb851b6d86c7eddc6e88a3f32027b1f0d","tag":"autogenerated:gerrit:merged","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2026-01-19 13:31:08.000000000","message":"Change has been successfully merged","accounts_in_message":[],"_revision_number":9},{"id":"30624bea89164d9a5eb9c90fdab212adf98e60ea","tag":"autogenerated:zuul:promote","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2026-01-19 13:33:53.000000000","message":"Patch Set 9:\n\nBuild succeeded (promote pipeline).\nhttps://zuul.opendev.org/t/openstack/buildset/ab40ba95ae4f4e4c9a2a64c449d255d7\n\n- promote-openstack-tox-docs https://zuul.opendev.org/t/openstack/build/21f89671726e4e3f90b7250a125c67ef : SUCCESS in 43s","accounts_in_message":[],"_revision_number":9}],"current_revision_number":9,"current_revision":"1b0a3b93864f724ddb5a3e673bb50d6e91d6750d","revisions":{"6ebbaf407aa39f37438ce82d0bbbf47420d520af":{"kind":"REWORK","_number":1,"created":"2026-01-09 13:39:39.000000000","uploader":{"_account_id":37598,"name":"Ivan Anfimov","display_name":"Ivan Anfimov","email":"lazekteam@gmail.com","username":"anfimovir"},"ref":"refs/changes/70/972870/1","fetch":{"anonymous http":{"url":"https://review.opendev.org/openstack/openstack-ansible-lxc_hosts","ref":"refs/changes/70/972870/1","commands":{"Checkout":"git fetch https://review.opendev.org/openstack/openstack-ansible-lxc_hosts refs/changes/70/972870/1 \u0026\u0026 git checkout FETCH_HEAD","Cherry Pick":"git fetch https://review.opendev.org/openstack/openstack-ansible-lxc_hosts refs/changes/70/972870/1 \u0026\u0026 git cherry-pick FETCH_HEAD","Format Patch":"git fetch https://review.opendev.org/openstack/openstack-ansible-lxc_hosts refs/changes/70/972870/1 \u0026\u0026 git format-patch -1 --stdout FETCH_HEAD","Pull":"git pull https://review.opendev.org/openstack/openstack-ansible-lxc_hosts refs/changes/70/972870/1"}}},"commit":{"parents":[{"commit":"94cecce779ff4e56365b506aed8ace70aff7bfea","subject":"Add systemd-udev to LXC cache for EL","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/openstack-ansible-lxc_hosts/commit/94cecce779ff4e56365b506aed8ace70aff7bfea"}]}],"author":{"name":"Dmitriy Rabotyagov","email":"dmitriy.rabotyagov@cleura.com","date":"2026-01-07 23:05:31.000000000","tz":60},"committer":{"name":"Ivan Anfimov","email":"lazekteam@gmail.com","date":"2026-01-09 13:39:39.000000000","tz":0},"subject":"Do not install post-quantum GPG keys","message":"Do not install post-quantum GPG keys\n\nAt the moment rpm_key fails to process PQC GPG keys while attempting\nto check if gpg is installed [1].\n\nThus, rpm_key needs to start running sequoia instead of gnupg to verify\nPQ keys.\n\nUntil then, we avoid importing PQC keys.\n\n[1] https://github.com/ansible/ansible/blob/69afa45880c9c03009b178632dccc7c0ffc5b5fa/lib/ansible/modules/rpm_key.py#L240\n\nChange-Id: If176d905411869b55ff1751659f5e46963c9f037\nSigned-off-by: Dmitriy Rabotyagov \u003cdmitriy.rabotyagov@cleura.com\u003e\n","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/openstack-ansible-lxc_hosts/commit/6ebbaf407aa39f37438ce82d0bbbf47420d520af"}],"resolve_conflicts_web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/openstack-ansible-lxc_hosts/commit/6ebbaf407aa39f37438ce82d0bbbf47420d520af"}]},"branch":"refs/heads/stable/2025.1"},"97268f79c0f18728541f7ae3a02f69abf839a460":{"kind":"TRIVIAL_REBASE","_number":2,"created":"2026-01-09 13:43:03.000000000","uploader":{"_account_id":37598,"name":"Ivan Anfimov","display_name":"Ivan Anfimov","email":"lazekteam@gmail.com","username":"anfimovir"},"ref":"refs/changes/70/972870/2","fetch":{"anonymous http":{"url":"https://review.opendev.org/openstack/openstack-ansible-lxc_hosts","ref":"refs/changes/70/972870/2","commands":{"Checkout":"git fetch https://review.opendev.org/openstack/openstack-ansible-lxc_hosts refs/changes/70/972870/2 \u0026\u0026 git checkout FETCH_HEAD","Cherry Pick":"git fetch https://review.opendev.org/openstack/openstack-ansible-lxc_hosts refs/changes/70/972870/2 \u0026\u0026 git cherry-pick FETCH_HEAD","Format Patch":"git fetch https://review.opendev.org/openstack/openstack-ansible-lxc_hosts refs/changes/70/972870/2 \u0026\u0026 git format-patch -1 --stdout FETCH_HEAD","Pull":"git pull https://review.opendev.org/openstack/openstack-ansible-lxc_hosts refs/changes/70/972870/2"}}},"commit":{"parents":[{"commit":"ccadfa55a5995813db034fe44a30607f9cdd27ac","subject":"Temporary disable RHEL jobs","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/openstack-ansible-lxc_hosts/commit/ccadfa55a5995813db034fe44a30607f9cdd27ac"}]}],"author":{"name":"Dmitriy Rabotyagov","email":"dmitriy.rabotyagov@cleura.com","date":"2026-01-07 23:05:31.000000000","tz":60},"committer":{"name":"Ivan Anfimov","email":"lazekteam@gmail.com","date":"2026-01-09 13:43:03.000000000","tz":0},"subject":"Do not install post-quantum GPG keys","message":"Do not install post-quantum GPG keys\n\nAt the moment rpm_key fails to process PQC GPG keys while attempting\nto check if gpg is installed [1].\n\nThus, rpm_key needs to start running sequoia instead of gnupg to verify\nPQ keys.\n\nUntil then, we avoid importing PQC keys.\n\n[1] https://github.com/ansible/ansible/blob/69afa45880c9c03009b178632dccc7c0ffc5b5fa/lib/ansible/modules/rpm_key.py#L240\n\nChange-Id: If176d905411869b55ff1751659f5e46963c9f037\nSigned-off-by: Dmitriy Rabotyagov \u003cdmitriy.rabotyagov@cleura.com\u003e\n","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/openstack-ansible-lxc_hosts/commit/97268f79c0f18728541f7ae3a02f69abf839a460"}],"resolve_conflicts_web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/openstack-ansible-lxc_hosts/commit/97268f79c0f18728541f7ae3a02f69abf839a460"}]},"branch":"refs/heads/stable/2025.1","description":"Rebase"},"454a0dcf19f7af427592efcf0af00c6e64c6c6e9":{"kind":"TRIVIAL_REBASE","_number":3,"created":"2026-01-09 15:44:34.000000000","uploader":{"_account_id":37598,"name":"Ivan Anfimov","display_name":"Ivan Anfimov","email":"lazekteam@gmail.com","username":"anfimovir"},"ref":"refs/changes/70/972870/3","fetch":{"anonymous http":{"url":"https://review.opendev.org/openstack/openstack-ansible-lxc_hosts","ref":"refs/changes/70/972870/3","commands":{"Checkout":"git fetch https://review.opendev.org/openstack/openstack-ansible-lxc_hosts refs/changes/70/972870/3 \u0026\u0026 git checkout FETCH_HEAD","Cherry Pick":"git fetch https://review.opendev.org/openstack/openstack-ansible-lxc_hosts refs/changes/70/972870/3 \u0026\u0026 git cherry-pick FETCH_HEAD","Format Patch":"git fetch https://review.opendev.org/openstack/openstack-ansible-lxc_hosts refs/changes/70/972870/3 \u0026\u0026 git format-patch -1 --stdout FETCH_HEAD","Pull":"git pull https://review.opendev.org/openstack/openstack-ansible-lxc_hosts refs/changes/70/972870/3"}}},"commit":{"parents":[{"commit":"f7ad113165f3e9474e3172680a7ab7d81bde88b3","subject":"Temporary disable RHEL jobs","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/openstack-ansible-lxc_hosts/commit/f7ad113165f3e9474e3172680a7ab7d81bde88b3"}]}],"author":{"name":"Dmitriy Rabotyagov","email":"dmitriy.rabotyagov@cleura.com","date":"2026-01-07 23:05:31.000000000","tz":60},"committer":{"name":"Ivan Anfimov","email":"lazekteam@gmail.com","date":"2026-01-09 15:44:34.000000000","tz":0},"subject":"Do not install post-quantum GPG keys","message":"Do not install post-quantum GPG keys\n\nAt the moment rpm_key fails to process PQC GPG keys while attempting\nto check if gpg is installed [1].\n\nThus, rpm_key needs to start running sequoia instead of gnupg to verify\nPQ keys.\n\nUntil then, we avoid importing PQC keys.\n\n[1] https://github.com/ansible/ansible/blob/69afa45880c9c03009b178632dccc7c0ffc5b5fa/lib/ansible/modules/rpm_key.py#L240\n\nChange-Id: If176d905411869b55ff1751659f5e46963c9f037\nSigned-off-by: Dmitriy Rabotyagov \u003cdmitriy.rabotyagov@cleura.com\u003e\n","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/openstack-ansible-lxc_hosts/commit/454a0dcf19f7af427592efcf0af00c6e64c6c6e9"}],"resolve_conflicts_web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/openstack-ansible-lxc_hosts/commit/454a0dcf19f7af427592efcf0af00c6e64c6c6e9"}]},"branch":"refs/heads/stable/2025.1","description":"Rebase"},"d8749d0c1ff92677379fe8cf3c3fe29e6f07bc6f":{"kind":"TRIVIAL_REBASE","_number":4,"created":"2026-01-09 17:39:58.000000000","uploader":{"_account_id":37598,"name":"Ivan Anfimov","display_name":"Ivan Anfimov","email":"lazekteam@gmail.com","username":"anfimovir"},"ref":"refs/changes/70/972870/4","fetch":{"anonymous http":{"url":"https://review.opendev.org/openstack/openstack-ansible-lxc_hosts","ref":"refs/changes/70/972870/4","commands":{"Checkout":"git fetch https://review.opendev.org/openstack/openstack-ansible-lxc_hosts refs/changes/70/972870/4 \u0026\u0026 git checkout FETCH_HEAD","Cherry Pick":"git fetch https://review.opendev.org/openstack/openstack-ansible-lxc_hosts refs/changes/70/972870/4 \u0026\u0026 git cherry-pick FETCH_HEAD","Format Patch":"git fetch https://review.opendev.org/openstack/openstack-ansible-lxc_hosts refs/changes/70/972870/4 \u0026\u0026 git format-patch -1 --stdout FETCH_HEAD","Pull":"git pull https://review.opendev.org/openstack/openstack-ansible-lxc_hosts refs/changes/70/972870/4"}}},"commit":{"parents":[{"commit":"79920cc18dfa642e0d60dab1ca0ccca05fe0d6ee","subject":"Temporary disable RHEL jobs","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/openstack-ansible-lxc_hosts/commit/79920cc18dfa642e0d60dab1ca0ccca05fe0d6ee"}]}],"author":{"name":"Dmitriy Rabotyagov","email":"dmitriy.rabotyagov@cleura.com","date":"2026-01-07 23:05:31.000000000","tz":60},"committer":{"name":"Ivan Anfimov","email":"lazekteam@gmail.com","date":"2026-01-09 17:39:58.000000000","tz":0},"subject":"Do not install post-quantum GPG keys","message":"Do not install post-quantum GPG keys\n\nAt the moment rpm_key fails to process PQC GPG keys while attempting\nto check if gpg is installed [1].\n\nThus, rpm_key needs to start running sequoia instead of gnupg to verify\nPQ keys.\n\nUntil then, we avoid importing PQC keys.\n\n[1] https://github.com/ansible/ansible/blob/69afa45880c9c03009b178632dccc7c0ffc5b5fa/lib/ansible/modules/rpm_key.py#L240\n\nChange-Id: If176d905411869b55ff1751659f5e46963c9f037\nSigned-off-by: Dmitriy Rabotyagov \u003cdmitriy.rabotyagov@cleura.com\u003e\n","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/openstack-ansible-lxc_hosts/commit/d8749d0c1ff92677379fe8cf3c3fe29e6f07bc6f"}],"resolve_conflicts_web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/openstack-ansible-lxc_hosts/commit/d8749d0c1ff92677379fe8cf3c3fe29e6f07bc6f"}]},"branch":"refs/heads/stable/2025.1","description":"Rebase"},"d6df6f2fb5f17e74de4ddf37591105151b35a145":{"kind":"NO_CODE_CHANGE","_number":5,"created":"2026-01-09 19:44:06.000000000","uploader":{"_account_id":37598,"name":"Ivan Anfimov","display_name":"Ivan Anfimov","email":"lazekteam@gmail.com","username":"anfimovir"},"ref":"refs/changes/70/972870/5","fetch":{"anonymous http":{"url":"https://review.opendev.org/openstack/openstack-ansible-lxc_hosts","ref":"refs/changes/70/972870/5","commands":{"Checkout":"git fetch https://review.opendev.org/openstack/openstack-ansible-lxc_hosts refs/changes/70/972870/5 \u0026\u0026 git checkout FETCH_HEAD","Cherry Pick":"git fetch https://review.opendev.org/openstack/openstack-ansible-lxc_hosts refs/changes/70/972870/5 \u0026\u0026 git cherry-pick FETCH_HEAD","Format Patch":"git fetch https://review.opendev.org/openstack/openstack-ansible-lxc_hosts refs/changes/70/972870/5 \u0026\u0026 git format-patch -1 --stdout FETCH_HEAD","Pull":"git pull https://review.opendev.org/openstack/openstack-ansible-lxc_hosts refs/changes/70/972870/5"}}},"commit":{"parents":[{"commit":"79920cc18dfa642e0d60dab1ca0ccca05fe0d6ee","subject":"Temporary disable RHEL jobs","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/openstack-ansible-lxc_hosts/commit/79920cc18dfa642e0d60dab1ca0ccca05fe0d6ee"}]}],"author":{"name":"Dmitriy Rabotyagov","email":"dmitriy.rabotyagov@cleura.com","date":"2026-01-07 23:05:31.000000000","tz":60},"committer":{"name":"Ivan Anfimov","email":"lazekteam@gmail.com","date":"2026-01-09 19:44:06.000000000","tz":0},"subject":"Do not install post-quantum GPG keys","message":"Do not install post-quantum GPG keys\n\nAt the moment rpm_key fails to process PQC GPG keys while attempting\nto check if gpg is installed [1].\n\nThus, rpm_key needs to start running sequoia instead of gnupg to verify\nPQ keys.\n\nUntil then, we avoid importing PQC keys.\n\n[1] https://github.com/ansible/ansible/blob/69afa45880c9c03009b178632dccc7c0ffc5b5fa/lib/ansible/modules/rpm_key.py#L240\n\nDepends-On: https://review.opendev.org/c/openstack/openstack-ansible/+/971749\nChange-Id: If176d905411869b55ff1751659f5e46963c9f037\nSigned-off-by: Dmitriy Rabotyagov \u003cdmitriy.rabotyagov@cleura.com\u003e\n","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/openstack-ansible-lxc_hosts/commit/d6df6f2fb5f17e74de4ddf37591105151b35a145"}],"resolve_conflicts_web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/openstack-ansible-lxc_hosts/commit/d6df6f2fb5f17e74de4ddf37591105151b35a145"}]},"branch":"refs/heads/stable/2025.1","description":"Edit commit message"},"c6befa49cdcac9186d4c3b2f6c6d8f5ca742f84e":{"kind":"NO_CHANGE","_number":6,"created":"2026-01-09 19:45:55.000000000","uploader":{"_account_id":37598,"name":"Ivan Anfimov","display_name":"Ivan Anfimov","email":"lazekteam@gmail.com","username":"anfimovir"},"ref":"refs/changes/70/972870/6","fetch":{"anonymous http":{"url":"https://review.opendev.org/openstack/openstack-ansible-lxc_hosts","ref":"refs/changes/70/972870/6","commands":{"Checkout":"git fetch https://review.opendev.org/openstack/openstack-ansible-lxc_hosts refs/changes/70/972870/6 \u0026\u0026 git checkout FETCH_HEAD","Cherry Pick":"git fetch https://review.opendev.org/openstack/openstack-ansible-lxc_hosts refs/changes/70/972870/6 \u0026\u0026 git cherry-pick FETCH_HEAD","Format Patch":"git fetch https://review.opendev.org/openstack/openstack-ansible-lxc_hosts refs/changes/70/972870/6 \u0026\u0026 git format-patch -1 --stdout FETCH_HEAD","Pull":"git pull https://review.opendev.org/openstack/openstack-ansible-lxc_hosts refs/changes/70/972870/6"}}},"commit":{"parents":[{"commit":"49e3304945b5f3094bc944924e4db7a161e0e9d2","subject":"Temporary disable RHEL jobs","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/openstack-ansible-lxc_hosts/commit/49e3304945b5f3094bc944924e4db7a161e0e9d2"}]}],"author":{"name":"Dmitriy Rabotyagov","email":"dmitriy.rabotyagov@cleura.com","date":"2026-01-07 23:05:31.000000000","tz":60},"committer":{"name":"Ivan Anfimov","email":"lazekteam@gmail.com","date":"2026-01-09 19:45:55.000000000","tz":0},"subject":"Do not install post-quantum GPG keys","message":"Do not install post-quantum GPG keys\n\nAt the moment rpm_key fails to process PQC GPG keys while attempting\nto check if gpg is installed [1].\n\nThus, rpm_key needs to start running sequoia instead of gnupg to verify\nPQ keys.\n\nUntil then, we avoid importing PQC keys.\n\n[1] https://github.com/ansible/ansible/blob/69afa45880c9c03009b178632dccc7c0ffc5b5fa/lib/ansible/modules/rpm_key.py#L240\n\nDepends-On: https://review.opendev.org/c/openstack/openstack-ansible/+/971749\nChange-Id: If176d905411869b55ff1751659f5e46963c9f037\nSigned-off-by: Dmitriy Rabotyagov \u003cdmitriy.rabotyagov@cleura.com\u003e\n","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/openstack-ansible-lxc_hosts/commit/c6befa49cdcac9186d4c3b2f6c6d8f5ca742f84e"}],"resolve_conflicts_web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/openstack-ansible-lxc_hosts/commit/c6befa49cdcac9186d4c3b2f6c6d8f5ca742f84e"}]},"branch":"refs/heads/stable/2025.1","description":"Rebase"},"db4bef39722a0643ee35c02c256683fa5cfe72c5":{"kind":"TRIVIAL_REBASE_WITH_MESSAGE_UPDATE","_number":7,"created":"2026-01-16 15:10:46.000000000","uploader":{"_account_id":28619,"name":"Dmitriy Rabotyagov","email":"noonedeadpunk@gmail.com","username":"noonedeadpunk"},"ref":"refs/changes/70/972870/7","fetch":{"anonymous http":{"url":"https://review.opendev.org/openstack/openstack-ansible-lxc_hosts","ref":"refs/changes/70/972870/7","commands":{"Checkout":"git fetch https://review.opendev.org/openstack/openstack-ansible-lxc_hosts refs/changes/70/972870/7 \u0026\u0026 git checkout FETCH_HEAD","Cherry Pick":"git fetch https://review.opendev.org/openstack/openstack-ansible-lxc_hosts refs/changes/70/972870/7 \u0026\u0026 git cherry-pick FETCH_HEAD","Format Patch":"git fetch https://review.opendev.org/openstack/openstack-ansible-lxc_hosts refs/changes/70/972870/7 \u0026\u0026 git format-patch -1 --stdout FETCH_HEAD","Pull":"git pull https://review.opendev.org/openstack/openstack-ansible-lxc_hosts refs/changes/70/972870/7"}}},"commit":{"parents":[{"commit":"94cecce779ff4e56365b506aed8ace70aff7bfea","subject":"Add systemd-udev to LXC cache for EL","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/openstack-ansible-lxc_hosts/commit/94cecce779ff4e56365b506aed8ace70aff7bfea"}]}],"author":{"name":"Dmitriy Rabotyagov","email":"dmitriy.rabotyagov@cleura.com","date":"2026-01-07 23:05:31.000000000","tz":60},"committer":{"name":"Dmitriy Rabotyagov","email":"dmitriy.rabotyagov@cleura.com","date":"2026-01-16 15:10:46.000000000","tz":0},"subject":"Do not install post-quantum GPG keys","message":"Do not install post-quantum GPG keys\n\nAt the moment rpm_key fails to process PQC GPG keys while attempting\nto check if gpg is installed [1].\n\nThus, rpm_key needs to start running sequoia instead of gnupg to verify\nPQ keys.\n\nUntil then, we avoid importing PQC keys.\n\n[1] https://github.com/ansible/ansible/blob/69afa45880c9c03009b178632dccc7c0ffc5b5fa/lib/ansible/modules/rpm_key.py#L240\n\nChange-Id: If176d905411869b55ff1751659f5e46963c9f037\nSigned-off-by: Dmitriy Rabotyagov \u003cdmitriy.rabotyagov@cleura.com\u003e\n","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/openstack-ansible-lxc_hosts/commit/db4bef39722a0643ee35c02c256683fa5cfe72c5"}],"resolve_conflicts_web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/openstack-ansible-lxc_hosts/commit/db4bef39722a0643ee35c02c256683fa5cfe72c5"}]},"branch":"refs/heads/stable/2025.1"},"138eb032a5a6fd9e728e17ba4502005b08c1be34":{"kind":"TRIVIAL_REBASE","_number":8,"created":"2026-01-16 15:10:59.000000000","uploader":{"_account_id":28619,"name":"Dmitriy Rabotyagov","email":"noonedeadpunk@gmail.com","username":"noonedeadpunk"},"ref":"refs/changes/70/972870/8","fetch":{"anonymous http":{"url":"https://review.opendev.org/openstack/openstack-ansible-lxc_hosts","ref":"refs/changes/70/972870/8","commands":{"Checkout":"git fetch https://review.opendev.org/openstack/openstack-ansible-lxc_hosts refs/changes/70/972870/8 \u0026\u0026 git checkout FETCH_HEAD","Cherry Pick":"git fetch https://review.opendev.org/openstack/openstack-ansible-lxc_hosts refs/changes/70/972870/8 \u0026\u0026 git cherry-pick FETCH_HEAD","Format Patch":"git fetch https://review.opendev.org/openstack/openstack-ansible-lxc_hosts refs/changes/70/972870/8 \u0026\u0026 git format-patch -1 --stdout FETCH_HEAD","Pull":"git pull https://review.opendev.org/openstack/openstack-ansible-lxc_hosts refs/changes/70/972870/8"}}},"commit":{"parents":[{"commit":"de38723455cbbdbb3f42b1373dc66400764a6df9","subject":"Temporary disable RHEL jobs","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/openstack-ansible-lxc_hosts/commit/de38723455cbbdbb3f42b1373dc66400764a6df9"}]}],"author":{"name":"Dmitriy Rabotyagov","email":"dmitriy.rabotyagov@cleura.com","date":"2026-01-07 23:05:31.000000000","tz":60},"committer":{"name":"Dmitriy Rabotyagov","email":"dmitriy.rabotyagov@cleura.com","date":"2026-01-16 15:10:59.000000000","tz":0},"subject":"Do not install post-quantum GPG keys","message":"Do not install post-quantum GPG keys\n\nAt the moment rpm_key fails to process PQC GPG keys while attempting\nto check if gpg is installed [1].\n\nThus, rpm_key needs to start running sequoia instead of gnupg to verify\nPQ keys.\n\nUntil then, we avoid importing PQC keys.\n\n[1] https://github.com/ansible/ansible/blob/69afa45880c9c03009b178632dccc7c0ffc5b5fa/lib/ansible/modules/rpm_key.py#L240\n\nChange-Id: If176d905411869b55ff1751659f5e46963c9f037\nSigned-off-by: Dmitriy Rabotyagov \u003cdmitriy.rabotyagov@cleura.com\u003e\n","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/openstack-ansible-lxc_hosts/commit/138eb032a5a6fd9e728e17ba4502005b08c1be34"}],"resolve_conflicts_web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/openstack-ansible-lxc_hosts/commit/138eb032a5a6fd9e728e17ba4502005b08c1be34"}]},"branch":"refs/heads/stable/2025.1","description":"Rebase"},"1b0a3b93864f724ddb5a3e673bb50d6e91d6750d":{"kind":"TRIVIAL_REBASE","_number":9,"created":"2026-01-16 17:08:22.000000000","uploader":{"_account_id":28619,"name":"Dmitriy Rabotyagov","email":"noonedeadpunk@gmail.com","username":"noonedeadpunk"},"ref":"refs/changes/70/972870/9","fetch":{"anonymous http":{"url":"https://review.opendev.org/openstack/openstack-ansible-lxc_hosts","ref":"refs/changes/70/972870/9","commands":{"Checkout":"git fetch https://review.opendev.org/openstack/openstack-ansible-lxc_hosts refs/changes/70/972870/9 \u0026\u0026 git checkout FETCH_HEAD","Cherry Pick":"git fetch https://review.opendev.org/openstack/openstack-ansible-lxc_hosts refs/changes/70/972870/9 \u0026\u0026 git cherry-pick FETCH_HEAD","Format Patch":"git fetch https://review.opendev.org/openstack/openstack-ansible-lxc_hosts refs/changes/70/972870/9 \u0026\u0026 git format-patch -1 --stdout FETCH_HEAD","Pull":"git pull https://review.opendev.org/openstack/openstack-ansible-lxc_hosts refs/changes/70/972870/9"}}},"commit":{"parents":[{"commit":"a8d047b8e1105080a588ed91b75e4ae537ec23de","subject":"Temporary disable RHEL jobs","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/openstack-ansible-lxc_hosts/commit/a8d047b8e1105080a588ed91b75e4ae537ec23de"}]}],"author":{"name":"Dmitriy Rabotyagov","email":"dmitriy.rabotyagov@cleura.com","date":"2026-01-07 23:05:31.000000000","tz":60},"committer":{"name":"Dmitriy Rabotyagov","email":"dmitriy.rabotyagov@cleura.com","date":"2026-01-16 17:08:22.000000000","tz":0},"subject":"Do not install post-quantum GPG keys","message":"Do not install post-quantum GPG keys\n\nAt the moment rpm_key fails to process PQC GPG keys while attempting\nto check if gpg is installed [1].\n\nThus, rpm_key needs to start running sequoia instead of gnupg to verify\nPQ keys.\n\nUntil then, we avoid importing PQC keys.\n\n[1] https://github.com/ansible/ansible/blob/69afa45880c9c03009b178632dccc7c0ffc5b5fa/lib/ansible/modules/rpm_key.py#L240\n\nChange-Id: If176d905411869b55ff1751659f5e46963c9f037\nSigned-off-by: Dmitriy Rabotyagov \u003cdmitriy.rabotyagov@cleura.com\u003e\n","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/openstack-ansible-lxc_hosts/commit/1b0a3b93864f724ddb5a3e673bb50d6e91d6750d"}],"resolve_conflicts_web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/openstack-ansible-lxc_hosts/commit/1b0a3b93864f724ddb5a3e673bb50d6e91d6750d"}]},"branch":"refs/heads/stable/2025.1","description":"Rebase"}},"requirements":[],"submit_records":[{"rule_name":"gerrit~DefaultSubmitRule","status":"CLOSED","labels":[{"label":"Verified","status":"MAY","applied_by":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]}},{"label":"Code-Review","status":"MAY","applied_by":{"_account_id":28619,"name":"Dmitriy Rabotyagov","email":"noonedeadpunk@gmail.com","username":"noonedeadpunk"}},{"label":"Workflow","status":"MAY","applied_by":{"_account_id":28619,"name":"Dmitriy Rabotyagov","email":"noonedeadpunk@gmail.com","username":"noonedeadpunk"}},{"label":"Backport-Candidate","status":"MAY"}]}],"submit_requirements":[{"name":"Verified","description":"Verified in gate by CI","status":"SATISFIED","is_legacy":false,"submittability_expression_result":{"expression":"label:Verified\u003dMAX AND -label:Verified\u003dMIN","fulfilled":true,"status":"PASS","passing_atoms":["label:Verified\u003dMAX"],"failing_atoms":["label:Verified\u003dMIN"],"atom_explanations":{}}},{"name":"Backport-Candidate","description":"Backport candidate status","status":"NOT_APPLICABLE","is_legacy":false,"applicability_expression_result":{"fulfilled":false,"status":"FAIL"},"submittability_expression_result":{"expression":"is:true","fulfilled":true,"status":"NOT_EVALUATED","passing_atoms":[],"failing_atoms":[],"atom_explanations":{}}},{"name":"Code-Review","description":"Code reviewed by core reviewer","status":"SATISFIED","is_legacy":false,"submittability_expression_result":{"expression":"label:Code-Review\u003dMAX AND -label:Code-Review\u003dMIN","fulfilled":true,"status":"PASS","passing_atoms":["label:Code-Review\u003dMAX"],"failing_atoms":["label:Code-Review\u003dMIN"],"atom_explanations":{}}},{"name":"Workflow","description":"Approved for gate by core reviewer","status":"SATISFIED","is_legacy":false,"submittability_expression_result":{"expression":"label:Workflow\u003dMAX AND -label:Workflow\u003dMIN","fulfilled":true,"status":"PASS","passing_atoms":["label:Workflow\u003dMAX"],"failing_atoms":["label:Workflow\u003dMIN"],"atom_explanations":{}}}]}
