)]}'
{"id":"openstack%2Fopenstack-ansible-lxc_hosts~972871","triplet_id":"openstack%2Fopenstack-ansible-lxc_hosts~stable%2F2025.2~Ic91f5275f2579b8a9a4eb79858fdf4fc368f61c9","project":"openstack/openstack-ansible-lxc_hosts","branch":"stable/2025.2","attention_set":{},"removed_from_attention_set":{"31542":{"account":{"_account_id":31542,"name":"Andrew Bonney","email":"andrew.bonney@bbc.co.uk","username":"andrewbonney"},"last_update":"2026-01-12 08:18:36.000000000","reason":"\u003cGERRIT_ACCOUNT_31542\u003e replied on the change","reason_account":{"_account_id":31542,"name":"Andrew Bonney","email":"andrew.bonney@bbc.co.uk","username":"andrewbonney"}},"32666":{"account":{"_account_id":32666,"name":"Damian Dąbrowski","email":"damian@dabrowski.cloud","username":"ddabrowski"},"last_update":"2026-01-19 20:40:57.000000000","reason":"Change was submitted"},"28619":{"account":{"_account_id":28619,"name":"Dmitriy Rabotyagov","email":"noonedeadpunk@gmail.com","username":"noonedeadpunk"},"last_update":"2026-01-19 18:41:39.000000000","reason":"\u003cGERRIT_ACCOUNT_28619\u003e replied on the change","reason_account":{"_account_id":28619,"name":"Dmitriy Rabotyagov","email":"noonedeadpunk@gmail.com","username":"noonedeadpunk"}},"37598":{"account":{"_account_id":37598,"name":"Ivan Anfimov","display_name":"Ivan Anfimov","email":"lazekteam@gmail.com","username":"anfimovir"},"last_update":"2026-01-19 20:40:57.000000000","reason":"Change was submitted"},"25023":{"account":{"_account_id":25023,"name":"Jonathan Rosser","email":"jonathan.rosser@rd.bbc.co.uk","username":"jrosser"},"last_update":"2026-01-19 20:40:57.000000000","reason":"Change was submitted"}},"hashtags":[],"change_id":"Ic91f5275f2579b8a9a4eb79858fdf4fc368f61c9","subject":"Temporary disable RHEL jobs","status":"MERGED","created":"2026-01-09 13:39:57.000000000","updated":"2026-01-19 20:42:43.000000000","submitted":"2026-01-19 20:40:57.000000000","submitter":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"total_comment_count":0,"unresolved_comment_count":0,"has_review_started":true,"submission_id":"972871","cherry_pick_of_change":972789,"cherry_pick_of_patch_set":2,"meta_rev_id":"f3bd57174782ea3b5a779df0fdd59ef284b2aab3","_number":972871,"virtual_id_number":972871,"owner":{"_account_id":37598,"name":"Ivan Anfimov","display_name":"Ivan Anfimov","email":"lazekteam@gmail.com","username":"anfimovir"},"actions":{},"labels":{"Verified":{"approved":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"all":[{"value":0,"_account_id":31542,"name":"Andrew Bonney","email":"andrew.bonney@bbc.co.uk","username":"andrewbonney"},{"value":0,"_account_id":28619,"name":"Dmitriy Rabotyagov","email":"noonedeadpunk@gmail.com","username":"noonedeadpunk"},{"tag":"autogenerated:zuul:gate","value":2,"date":"2026-01-19 20:40:57.000000000","permitted_voting_range":{"min":2,"max":2},"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]}],"values":{"-2":"Fails","-1":"Doesn\u0027t seem to work"," 0":"No score","+1":"Works for me","+2":"Verified"},"description":"","default_value":0,"optional":true},"Code-Review":{"approved":{"_account_id":31542,"name":"Andrew Bonney","email":"andrew.bonney@bbc.co.uk","username":"andrewbonney"},"all":[{"value":2,"date":"2026-01-12 08:18:36.000000000","permitted_voting_range":{"min":2,"max":2},"_account_id":31542,"name":"Andrew Bonney","email":"andrew.bonney@bbc.co.uk","username":"andrewbonney"},{"value":2,"date":"2026-01-19 18:41:39.000000000","permitted_voting_range":{"min":2,"max":2},"_account_id":28619,"name":"Dmitriy Rabotyagov","email":"noonedeadpunk@gmail.com","username":"noonedeadpunk"},{"value":0,"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]}],"values":{"-2":"Do not merge","-1":"This patch needs further work before it can be merged"," 0":"No score","+1":"Looks good to me, but someone else must approve","+2":"Looks good to me (core reviewer)"},"description":"","default_value":0,"optional":true},"Workflow":{"approved":{"_account_id":28619,"name":"Dmitriy Rabotyagov","email":"noonedeadpunk@gmail.com","username":"noonedeadpunk"},"all":[{"value":0,"_account_id":31542,"name":"Andrew Bonney","email":"andrew.bonney@bbc.co.uk","username":"andrewbonney"},{"value":1,"date":"2026-01-19 18:41:39.000000000","permitted_voting_range":{"min":1,"max":1},"_account_id":28619,"name":"Dmitriy Rabotyagov","email":"noonedeadpunk@gmail.com","username":"noonedeadpunk"},{"value":0,"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]}],"values":{"-1":"Work in progress"," 0":"Ready for reviews","+1":"Approved"},"description":"","default_value":0,"optional":true},"Backport-Candidate":{"all":[{"value":0,"_account_id":31542,"name":"Andrew Bonney","email":"andrew.bonney@bbc.co.uk","username":"andrewbonney"},{"value":0,"_account_id":28619,"name":"Dmitriy Rabotyagov","email":"noonedeadpunk@gmail.com","username":"noonedeadpunk"},{"value":0,"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]}],"values":{"-1":"Not A Backport Candidate"," 0":"Backport Review Needed","+1":"Proposed Backport"},"description":"","default_value":0,"optional":true}},"removable_reviewers":[{"_account_id":32666,"name":"Damian Dąbrowski","email":"damian@dabrowski.cloud","username":"ddabrowski"},{"_account_id":25023,"name":"Jonathan Rosser","email":"jonathan.rosser@rd.bbc.co.uk","username":"jrosser"}],"reviewers":{"REVIEWER":[{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},{"_account_id":25023,"name":"Jonathan Rosser","email":"jonathan.rosser@rd.bbc.co.uk","username":"jrosser"},{"_account_id":28619,"name":"Dmitriy Rabotyagov","email":"noonedeadpunk@gmail.com","username":"noonedeadpunk"},{"_account_id":31542,"name":"Andrew Bonney","email":"andrew.bonney@bbc.co.uk","username":"andrewbonney"},{"_account_id":32666,"name":"Damian Dąbrowski","email":"damian@dabrowski.cloud","username":"ddabrowski"}]},"pending_reviewers":{},"reviewer_updates":[{"updated":"2026-01-09 13:44:47.000000000","updated_by":{"_account_id":37598,"name":"Ivan Anfimov","display_name":"Ivan Anfimov","email":"lazekteam@gmail.com","username":"anfimovir"},"reviewer":{"_account_id":31542,"name":"Andrew Bonney","email":"andrew.bonney@bbc.co.uk","username":"andrewbonney"},"state":"REVIEWER"},{"updated":"2026-01-09 13:44:47.000000000","updated_by":{"_account_id":37598,"name":"Ivan Anfimov","display_name":"Ivan Anfimov","email":"lazekteam@gmail.com","username":"anfimovir"},"reviewer":{"_account_id":25023,"name":"Jonathan Rosser","email":"jonathan.rosser@rd.bbc.co.uk","username":"jrosser"},"state":"REVIEWER"},{"updated":"2026-01-09 13:44:47.000000000","updated_by":{"_account_id":37598,"name":"Ivan Anfimov","display_name":"Ivan Anfimov","email":"lazekteam@gmail.com","username":"anfimovir"},"reviewer":{"_account_id":28619,"name":"Dmitriy Rabotyagov","email":"noonedeadpunk@gmail.com","username":"noonedeadpunk"},"state":"REVIEWER"},{"updated":"2026-01-09 16:23:09.000000000","updated_by":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"reviewer":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"state":"REVIEWER"},{"updated":"2026-01-09 19:46:33.000000000","updated_by":{"_account_id":37598,"name":"Ivan Anfimov","display_name":"Ivan Anfimov","email":"lazekteam@gmail.com","username":"anfimovir"},"reviewer":{"_account_id":32666,"name":"Damian Dąbrowski","email":"damian@dabrowski.cloud","username":"ddabrowski"},"state":"REVIEWER"}],"messages":[{"id":"798edd39ea29b17bac730eeda7358afd74785f08","tag":"autogenerated:gerrit:newPatchSet","author":{"_account_id":37598,"name":"Ivan Anfimov","display_name":"Ivan Anfimov","email":"lazekteam@gmail.com","username":"anfimovir"},"date":"2026-01-09 13:39:57.000000000","message":"Patch Set 1: Cherry Picked from branch master.","accounts_in_message":[],"_revision_number":1},{"id":"ada75effbae897d04b328ae8ba21f3f8de083e68","tag":"autogenerated:gerrit:newPatchSet","author":{"_account_id":37598,"name":"Ivan Anfimov","display_name":"Ivan Anfimov","email":"lazekteam@gmail.com","username":"anfimovir"},"date":"2026-01-09 13:41:11.000000000","message":"Patch Set 2: Commit message was updated.","accounts_in_message":[],"_revision_number":2},{"id":"663ff0f4c3bdb44a88c16b7ac6f72ce31f6d1b1e","tag":"autogenerated:zuul:check","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2026-01-09 16:23:09.000000000","message":"Patch Set 2: Verified+1\n\nBuild succeeded (check pipeline).\nhttps://zuul.opendev.org/t/openstack/buildset/e0214cf8dade4351a5e849a19cfa75df\n\n- openstack-ansible-linters-ubuntu-noble https://zuul.opendev.org/t/openstack/build/500d57a8d9a1422ea15ffbcb798a723f : SUCCESS in 10m 21s\n- openstack-tox-docs https://zuul.opendev.org/t/openstack/build/2b13d0189667429e8637a8d10cfd2bcf : SUCCESS in 6m 10s\n- openstack-ansible-deploy-infra_lxc-centos-10-stream https://zuul.opendev.org/t/openstack/build/14c4b74768df45a4840bd41002862ba2 : FAILURE in 22m 21s (non-voting)\n- openstack-ansible-deploy-infra_lxc-debian-bookworm https://zuul.opendev.org/t/openstack/build/f6b35582afbf4179a4008d0df501fbdc : SUCCESS in 56m 26s\n- openstack-ansible-deploy-infra_lxc-debian-trixie https://zuul.opendev.org/t/openstack/build/2d86b4b4bcf54f859c3efa01bc998e30 : SUCCESS in 1h 17m 59s\n- openstack-ansible-deploy-infra_lxc-rockylinux-10 https://zuul.opendev.org/t/openstack/build/732d7ac325fa4d76b23656c17bb1f26b : FAILURE in 56m 36s (non-voting)\n- openstack-ansible-deploy-infra_lxc_tls-rockylinux-10 https://zuul.opendev.org/t/openstack/build/ba713bd54a784e46ac8f1a6e10510d76 : FAILURE in 45m 52s (non-voting)\n- openstack-ansible-deploy-infra_lxc-ubuntu-noble https://zuul.opendev.org/t/openstack/build/db31013b4be940d8bcc2a066337d54be : SUCCESS in 1h 24m 53s\n- openstack-ansible-deploy-infra_lxc_stepca-ubuntu-noble https://zuul.opendev.org/t/openstack/build/7cf2f84a7a7d4c41afe2fd9a9d631ed3 : SUCCESS in 1h 19m 38s\n- openstack-ansible-deploy-infra_lxc_tls-ubuntu-noble https://zuul.opendev.org/t/openstack/build/4c9a7a48caf947b7b7026fa9a51f7b7d : SUCCESS in 1h 14m 12s\n- openstack-ansible-upgrade-infra_lxc-ubuntu-noble https://zuul.opendev.org/t/openstack/build/f3830f493ec644528b8abaad1a4bde30 : SUCCESS in 2h 11m 33s\n- openstack-ansible-deploy-hosts_lxc_zfs-ubuntu-noble https://zuul.opendev.org/t/openstack/build/44bc8e6f79cf469a900c4f2499ec80d3 : SUCCESS in 15m 21s\n- openstack-ansible-deploy-hosts_lxc_btrfs-ubuntu-noble https://zuul.opendev.org/t/openstack/build/e1beeb38e2d946bf84f11c313d376aca : SUCCESS in 28m 21s\n- openstack-ansible-deploy-hosts_lxc_overlayfs-ubuntu-noble https://zuul.opendev.org/t/openstack/build/cbcb94e363464305bd2bf02c30989ca0 : SUCCESS in 26m 59s\n- openstack-ansible-deploy-hosts_lxc_btrfs-debian-bookworm https://zuul.opendev.org/t/openstack/build/9cd37f7cb33f40729ad7be014e89b630 : SUCCESS in 27m 40s\n- openstack-ansible-deploy-hosts_lxc_overlayfs-debian-bookworm https://zuul.opendev.org/t/openstack/build/96a2c9db167d4bd9af78c45056d9d2fd : SUCCESS in 28m 49s\n- openstack-ansible-deploy-hosts_lxc_overlayfs-rockylinux-10 https://zuul.opendev.org/t/openstack/build/75c3c0ffec924b3a951b11fcb7f5f533 : SUCCESS in 20m 20s\n- openstack-ansible-deploy-hosts_lxc_btrfs-debian-trixie https://zuul.opendev.org/t/openstack/build/08817ef43df54cc18e7dcc53f818b2c6 : SUCCESS in 28m 08s\n- openstack-ansible-deploy-hosts_lxc_overlayfs-debian-trixie https://zuul.opendev.org/t/openstack/build/f10a1aaa7e014028bdc1eab1e9b818cf : SUCCESS in 20m 47s","accounts_in_message":[],"_revision_number":2},{"id":"31308af7c58c01a44c9c1c1878ed46a3917cfb3e","author":{"_account_id":31542,"name":"Andrew Bonney","email":"andrew.bonney@bbc.co.uk","username":"andrewbonney"},"date":"2026-01-12 08:18:36.000000000","message":"Patch Set 2: Code-Review+2","accounts_in_message":[],"_revision_number":2},{"id":"0cacdebb370584de53ac96d8eec200972edd8d3a","author":{"_account_id":28619,"name":"Dmitriy Rabotyagov","email":"noonedeadpunk@gmail.com","username":"noonedeadpunk"},"date":"2026-01-19 18:41:39.000000000","message":"Patch Set 2: Code-Review+2 Workflow+1","accounts_in_message":[],"_revision_number":2},{"id":"ccb0050efc03135273d6148bb242a83b11711f8f","tag":"autogenerated:zuul:gate","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2026-01-19 18:44:03.000000000","message":"Patch Set 2: -Verified\n\nStarting gate jobs.","accounts_in_message":[],"_revision_number":2},{"id":"c40ad29928e064430fdf064f2c7d444a2918d89d","tag":"autogenerated:zuul:gate","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2026-01-19 20:40:57.000000000","message":"Patch Set 2: Verified+2\n\nBuild succeeded (gate pipeline).\nhttps://zuul.opendev.org/t/openstack/buildset/50f14a6801a2479bbb93517de8a0719a\n\n- openstack-ansible-linters-ubuntu-noble https://zuul.opendev.org/t/openstack/build/e25cce747f1f445ebf1f7ef3c5dff71b : SUCCESS in 11m 20s\n- openstack-tox-docs https://zuul.opendev.org/t/openstack/build/ddd6ab6cec5d4a77b6cfa074f0abb41d : SUCCESS in 4m 10s\n- openstack-ansible-deploy-infra_lxc-centos-10-stream https://zuul.opendev.org/t/openstack/build/098c2f4cdc2c409ba7f8bbaa26a7ef8b : FAILURE in 24m 04s (non-voting)\n- openstack-ansible-deploy-infra_lxc-debian-bookworm https://zuul.opendev.org/t/openstack/build/f413aad83c06473db33efd0fc8d3545e : SUCCESS in 46m 52s\n- openstack-ansible-deploy-infra_lxc-debian-trixie https://zuul.opendev.org/t/openstack/build/1cb9ef43976f455b8288f562df8e4d0f : SUCCESS in 48m 31s\n- openstack-ansible-deploy-infra_lxc-rockylinux-10 https://zuul.opendev.org/t/openstack/build/03a5c95e3ece4844a8091dd9cfecdcdc : FAILURE in 29m 57s (non-voting)\n- openstack-ansible-deploy-infra_lxc_tls-rockylinux-10 https://zuul.opendev.org/t/openstack/build/3c1ac8203ef8488ca51c723d4bf11311 : FAILURE in 53m 38s (non-voting)\n- openstack-ansible-deploy-infra_lxc-ubuntu-noble https://zuul.opendev.org/t/openstack/build/c07bcf049955417b8615178fdf065c2c : SUCCESS in 44m 40s\n- openstack-ansible-deploy-infra_lxc_stepca-ubuntu-noble https://zuul.opendev.org/t/openstack/build/74569a26e1d2402d95db7fe28f820de2 : SUCCESS in 44m 07s\n- openstack-ansible-deploy-infra_lxc_tls-ubuntu-noble https://zuul.opendev.org/t/openstack/build/bdaf6ec261cc4fc9a306117c1024cce9 : SUCCESS in 1h 17m 03s\n- openstack-ansible-upgrade-infra_lxc-ubuntu-noble https://zuul.opendev.org/t/openstack/build/94347ba6ee064258ab0b0b8ead994e16 : SUCCESS in 1h 52m 06s\n- openstack-ansible-deploy-hosts_lxc_zfs-ubuntu-noble https://zuul.opendev.org/t/openstack/build/b9c6a4f47f944d6b8f4847f38beb5b6c : SUCCESS in 29m 35s\n- openstack-ansible-deploy-hosts_lxc_btrfs-ubuntu-noble https://zuul.opendev.org/t/openstack/build/1decf1aa7c694270b5ba40fbf48d2394 : SUCCESS in 19m 26s\n- openstack-ansible-deploy-hosts_lxc_overlayfs-ubuntu-noble https://zuul.opendev.org/t/openstack/build/36d0af1eeeeb464fa4b6712534bfe508 : SUCCESS in 18m 15s\n- openstack-ansible-deploy-hosts_lxc_btrfs-debian-bookworm https://zuul.opendev.org/t/openstack/build/caf26c0d7c124fa4956d154e78a256fa : SUCCESS in 24m 38s\n- openstack-ansible-deploy-hosts_lxc_overlayfs-debian-bookworm https://zuul.opendev.org/t/openstack/build/24f8fcecc1724e47b246e44e45c92201 : SUCCESS in 19m 46s\n- openstack-ansible-deploy-hosts_lxc_btrfs-debian-trixie https://zuul.opendev.org/t/openstack/build/17b1243f145740189a7b3a811533c659 : SUCCESS in 24m 48s\n- openstack-ansible-deploy-hosts_lxc_overlayfs-debian-trixie https://zuul.opendev.org/t/openstack/build/677e004048a14d0e9d1de55ddc888bef : SUCCESS in 25m 54s\n- openstack-ansible-deploy-hosts_lxc_overlayfs-rockylinux-10 https://zuul.opendev.org/t/openstack/build/a95565199338474e82984d19e15591c0 : SUCCESS in 32m 00s","accounts_in_message":[],"_revision_number":2},{"id":"125ebf42d361fa00b2ab13bfc33df35d37ef445a","tag":"autogenerated:gerrit:merged","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2026-01-19 20:40:57.000000000","message":"Change has been successfully merged","accounts_in_message":[],"_revision_number":2},{"id":"f3bd57174782ea3b5a779df0fdd59ef284b2aab3","tag":"autogenerated:zuul:promote","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2026-01-19 20:42:43.000000000","message":"Patch Set 2:\n\nBuild succeeded (promote pipeline).\nhttps://zuul.opendev.org/t/openstack/buildset/9f10a1f34a5e44d6bc2354d0135d9117\n\n- promote-openstack-tox-docs https://zuul.opendev.org/t/openstack/build/d7777a9a61604e3fb905c49ae1c214fb : SUCCESS in 45s","accounts_in_message":[],"_revision_number":2}],"current_revision_number":2,"current_revision":"b213e5012bfe538971cf77889c12128a5b9e3592","revisions":{"1fc4528ab59ab37bf2eb7329b764771f58ecc3e8":{"kind":"REWORK","_number":1,"created":"2026-01-09 13:39:57.000000000","uploader":{"_account_id":37598,"name":"Ivan Anfimov","display_name":"Ivan Anfimov","email":"lazekteam@gmail.com","username":"anfimovir"},"ref":"refs/changes/71/972871/1","fetch":{"anonymous http":{"url":"https://review.opendev.org/openstack/openstack-ansible-lxc_hosts","ref":"refs/changes/71/972871/1","commands":{"Checkout":"git fetch https://review.opendev.org/openstack/openstack-ansible-lxc_hosts refs/changes/71/972871/1 \u0026\u0026 git checkout FETCH_HEAD","Cherry Pick":"git fetch https://review.opendev.org/openstack/openstack-ansible-lxc_hosts refs/changes/71/972871/1 \u0026\u0026 git cherry-pick FETCH_HEAD","Format Patch":"git fetch https://review.opendev.org/openstack/openstack-ansible-lxc_hosts refs/changes/71/972871/1 \u0026\u0026 git format-patch -1 --stdout FETCH_HEAD","Pull":"git pull https://review.opendev.org/openstack/openstack-ansible-lxc_hosts refs/changes/71/972871/1"}}},"commit":{"parents":[{"commit":"52a68de3f3daa1f5954b1d78520aa395a1fd30cd","subject":"Add systemd-udev to LXC cache for EL","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/openstack-ansible-lxc_hosts/commit/52a68de3f3daa1f5954b1d78520aa395a1fd30cd"}]}],"author":{"name":"Ivan Anfimov","email":"lazekteam@gmail.com","date":"2026-01-08 20:42:18.000000000","tz":0},"committer":{"name":"Ivan Anfimov","email":"lazekteam@gmail.com","date":"2026-01-09 13:39:57.000000000","tz":0},"subject":"Temporary disable RHEL jobs","message":"Temporary disable RHEL jobs\n\nNeeded-By: https://review.opendev.org/c/openstack/openstack-ansible-lxc_hosts/+/972599\n\nChange-Id: Ic91f5275f2579b8a9a4eb79858fdf4fc368f61c9\nSigned-off-by: Ivan Anfimov \u003clazekteam@gmail.com\u003e\n","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/openstack-ansible-lxc_hosts/commit/1fc4528ab59ab37bf2eb7329b764771f58ecc3e8"}],"resolve_conflicts_web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/openstack-ansible-lxc_hosts/commit/1fc4528ab59ab37bf2eb7329b764771f58ecc3e8"}]},"branch":"refs/heads/stable/2025.2"},"b213e5012bfe538971cf77889c12128a5b9e3592":{"kind":"NO_CODE_CHANGE","_number":2,"created":"2026-01-09 13:41:11.000000000","uploader":{"_account_id":37598,"name":"Ivan Anfimov","display_name":"Ivan Anfimov","email":"lazekteam@gmail.com","username":"anfimovir"},"ref":"refs/changes/71/972871/2","fetch":{"anonymous http":{"url":"https://review.opendev.org/openstack/openstack-ansible-lxc_hosts","ref":"refs/changes/71/972871/2","commands":{"Checkout":"git fetch https://review.opendev.org/openstack/openstack-ansible-lxc_hosts refs/changes/71/972871/2 \u0026\u0026 git checkout FETCH_HEAD","Cherry Pick":"git fetch https://review.opendev.org/openstack/openstack-ansible-lxc_hosts refs/changes/71/972871/2 \u0026\u0026 git cherry-pick FETCH_HEAD","Format Patch":"git fetch https://review.opendev.org/openstack/openstack-ansible-lxc_hosts refs/changes/71/972871/2 \u0026\u0026 git format-patch -1 --stdout FETCH_HEAD","Pull":"git pull https://review.opendev.org/openstack/openstack-ansible-lxc_hosts refs/changes/71/972871/2"}}},"commit":{"parents":[{"commit":"52a68de3f3daa1f5954b1d78520aa395a1fd30cd","subject":"Add systemd-udev to LXC cache for EL","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/openstack-ansible-lxc_hosts/commit/52a68de3f3daa1f5954b1d78520aa395a1fd30cd"}]}],"author":{"name":"Ivan Anfimov","email":"lazekteam@gmail.com","date":"2026-01-08 20:42:18.000000000","tz":0},"committer":{"name":"Ivan Anfimov","email":"lazekteam@gmail.com","date":"2026-01-09 13:41:11.000000000","tz":0},"subject":"Temporary disable RHEL jobs","message":"Temporary disable RHEL jobs\n\nNeeded-By: https://review.opendev.org/c/openstack/openstack-ansible-lxc_hosts/+/972868\n\nChange-Id: Ic91f5275f2579b8a9a4eb79858fdf4fc368f61c9\nSigned-off-by: Ivan Anfimov \u003clazekteam@gmail.com\u003e\n","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/openstack-ansible-lxc_hosts/commit/b213e5012bfe538971cf77889c12128a5b9e3592"}],"resolve_conflicts_web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/openstack-ansible-lxc_hosts/commit/b213e5012bfe538971cf77889c12128a5b9e3592"}]},"branch":"refs/heads/stable/2025.2","description":"Edit commit message"}},"requirements":[],"submit_records":[{"rule_name":"gerrit~DefaultSubmitRule","status":"CLOSED","labels":[{"label":"Verified","status":"MAY","applied_by":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]}},{"label":"Code-Review","status":"MAY","applied_by":{"_account_id":28619,"name":"Dmitriy Rabotyagov","email":"noonedeadpunk@gmail.com","username":"noonedeadpunk"}},{"label":"Workflow","status":"MAY","applied_by":{"_account_id":28619,"name":"Dmitriy Rabotyagov","email":"noonedeadpunk@gmail.com","username":"noonedeadpunk"}},{"label":"Backport-Candidate","status":"MAY"}]}],"submit_requirements":[{"name":"Verified","description":"Verified in gate by CI","status":"SATISFIED","is_legacy":false,"submittability_expression_result":{"expression":"label:Verified\u003dMAX AND -label:Verified\u003dMIN","fulfilled":true,"status":"PASS","passing_atoms":["label:Verified\u003dMAX"],"failing_atoms":["label:Verified\u003dMIN"],"atom_explanations":{}}},{"name":"Backport-Candidate","description":"Backport candidate status","status":"NOT_APPLICABLE","is_legacy":false,"applicability_expression_result":{"fulfilled":false,"status":"FAIL"},"submittability_expression_result":{"expression":"is:true","fulfilled":true,"status":"NOT_EVALUATED","passing_atoms":[],"failing_atoms":[],"atom_explanations":{}}},{"name":"Code-Review","description":"Code reviewed by core reviewer","status":"SATISFIED","is_legacy":false,"submittability_expression_result":{"expression":"label:Code-Review\u003dMAX AND -label:Code-Review\u003dMIN","fulfilled":true,"status":"PASS","passing_atoms":["label:Code-Review\u003dMAX"],"failing_atoms":["label:Code-Review\u003dMIN"],"atom_explanations":{}}},{"name":"Workflow","description":"Approved for gate by core reviewer","status":"SATISFIED","is_legacy":false,"submittability_expression_result":{"expression":"label:Workflow\u003dMAX AND -label:Workflow\u003dMIN","fulfilled":true,"status":"PASS","passing_atoms":["label:Workflow\u003dMAX"],"failing_atoms":["label:Workflow\u003dMIN"],"atom_explanations":{}}}]}
