)]}'
{"id":"openstack%2Fopenstack-ansible-lxc_hosts~973887","triplet_id":"openstack%2Fopenstack-ansible-lxc_hosts~stable%2F2025.2~Ib1c557950c45276b26a8584bcae4891fed2aaa71","project":"openstack/openstack-ansible-lxc_hosts","branch":"stable/2025.2","attention_set":{},"removed_from_attention_set":{"31542":{"account":{"_account_id":31542,"name":"Andrew Bonney","email":"andrew.bonney@bbc.co.uk","username":"andrewbonney"},"last_update":"2026-01-20 08:23:45.000000000","reason":"\u003cGERRIT_ACCOUNT_31542\u003e replied on the change","reason_account":{"_account_id":31542,"name":"Andrew Bonney","email":"andrew.bonney@bbc.co.uk","username":"andrewbonney"}},"32666":{"account":{"_account_id":32666,"name":"Damian Dąbrowski","email":"damian@dabrowski.cloud","username":"ddabrowski"},"last_update":"2026-01-21 23:44:09.000000000","reason":"Change was submitted"},"28619":{"account":{"_account_id":28619,"name":"Dmitriy Rabotyagov","email":"noonedeadpunk@gmail.com","username":"noonedeadpunk"},"last_update":"2026-01-21 23:44:09.000000000","reason":"Change was submitted"},"25023":{"account":{"_account_id":25023,"name":"Jonathan Rosser","email":"jonathan.rosser@rd.bbc.co.uk","username":"jrosser"},"last_update":"2026-01-21 23:44:09.000000000","reason":"Change was submitted"}},"hashtags":[],"change_id":"Ib1c557950c45276b26a8584bcae4891fed2aaa71","subject":"Revert \"Temporary disable RHEL jobs\"","status":"MERGED","created":"2026-01-19 18:43:11.000000000","updated":"2026-01-21 23:45:33.000000000","submitted":"2026-01-21 23:44:09.000000000","submitter":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"total_comment_count":2,"unresolved_comment_count":0,"has_review_started":true,"submission_id":"973887","meta_rev_id":"cc31553a5046fadba2a1a14e9fa578d6831e51a9","_number":973887,"virtual_id_number":973887,"owner":{"_account_id":28619,"name":"Dmitriy Rabotyagov","email":"noonedeadpunk@gmail.com","username":"noonedeadpunk"},"actions":{},"labels":{"Verified":{"approved":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"all":[{"value":0,"_account_id":31542,"name":"Andrew Bonney","email":"andrew.bonney@bbc.co.uk","username":"andrewbonney"},{"value":0,"_account_id":28619,"name":"Dmitriy Rabotyagov","email":"noonedeadpunk@gmail.com","username":"noonedeadpunk"},{"tag":"autogenerated:zuul:gate","value":2,"date":"2026-01-21 23:44:09.000000000","permitted_voting_range":{"min":2,"max":2},"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},{"value":0,"_account_id":37598,"name":"Ivan Anfimov","display_name":"Ivan Anfimov","email":"lazekteam@gmail.com","username":"anfimovir"}],"values":{"-2":"Fails","-1":"Doesn\u0027t seem to work"," 0":"No score","+1":"Works for me","+2":"Verified"},"description":"","default_value":0,"optional":true},"Code-Review":{"approved":{"_account_id":31542,"name":"Andrew Bonney","email":"andrew.bonney@bbc.co.uk","username":"andrewbonney"},"all":[{"value":2,"date":"2026-01-20 08:23:45.000000000","permitted_voting_range":{"min":2,"max":2},"_account_id":31542,"name":"Andrew Bonney","email":"andrew.bonney@bbc.co.uk","username":"andrewbonney"},{"value":2,"date":"2026-01-21 14:57:42.000000000","permitted_voting_range":{"min":2,"max":2},"_account_id":28619,"name":"Dmitriy Rabotyagov","email":"noonedeadpunk@gmail.com","username":"noonedeadpunk"},{"value":0,"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},{"value":1,"date":"2026-01-19 19:42:18.000000000","permitted_voting_range":{"min":1,"max":2},"_account_id":37598,"name":"Ivan Anfimov","display_name":"Ivan Anfimov","email":"lazekteam@gmail.com","username":"anfimovir"}],"values":{"-2":"Do not merge","-1":"This patch needs further work before it can be merged"," 0":"No score","+1":"Looks good to me, but someone else must approve","+2":"Looks good to me (core reviewer)"},"description":"","default_value":0,"optional":true},"Workflow":{"approved":{"_account_id":28619,"name":"Dmitriy Rabotyagov","email":"noonedeadpunk@gmail.com","username":"noonedeadpunk"},"all":[{"value":0,"_account_id":31542,"name":"Andrew Bonney","email":"andrew.bonney@bbc.co.uk","username":"andrewbonney"},{"value":1,"date":"2026-01-21 14:57:42.000000000","permitted_voting_range":{"min":1,"max":1},"_account_id":28619,"name":"Dmitriy Rabotyagov","email":"noonedeadpunk@gmail.com","username":"noonedeadpunk"},{"value":0,"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},{"value":0,"_account_id":37598,"name":"Ivan Anfimov","display_name":"Ivan Anfimov","email":"lazekteam@gmail.com","username":"anfimovir"}],"values":{"-1":"Work in progress"," 0":"Ready for reviews","+1":"Approved"},"description":"","default_value":0,"optional":true},"Backport-Candidate":{"all":[{"value":0,"_account_id":31542,"name":"Andrew Bonney","email":"andrew.bonney@bbc.co.uk","username":"andrewbonney"},{"value":0,"_account_id":28619,"name":"Dmitriy Rabotyagov","email":"noonedeadpunk@gmail.com","username":"noonedeadpunk"},{"value":0,"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},{"value":0,"_account_id":37598,"name":"Ivan Anfimov","display_name":"Ivan Anfimov","email":"lazekteam@gmail.com","username":"anfimovir"}],"values":{"-1":"Not A Backport Candidate"," 0":"Backport Review Needed","+1":"Proposed Backport"},"description":"","default_value":0,"optional":true}},"removable_reviewers":[{"_account_id":32666,"name":"Damian Dąbrowski","email":"damian@dabrowski.cloud","username":"ddabrowski"},{"_account_id":25023,"name":"Jonathan Rosser","email":"jonathan.rosser@rd.bbc.co.uk","username":"jrosser"}],"reviewers":{"REVIEWER":[{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},{"_account_id":25023,"name":"Jonathan Rosser","email":"jonathan.rosser@rd.bbc.co.uk","username":"jrosser"},{"_account_id":28619,"name":"Dmitriy Rabotyagov","email":"noonedeadpunk@gmail.com","username":"noonedeadpunk"},{"_account_id":31542,"name":"Andrew Bonney","email":"andrew.bonney@bbc.co.uk","username":"andrewbonney"},{"_account_id":32666,"name":"Damian Dąbrowski","email":"damian@dabrowski.cloud","username":"ddabrowski"},{"_account_id":37598,"name":"Ivan Anfimov","display_name":"Ivan Anfimov","email":"lazekteam@gmail.com","username":"anfimovir"}]},"pending_reviewers":{},"reviewer_updates":[{"updated":"2026-01-19 18:43:11.000000000","updated_by":{"_account_id":28619,"name":"Dmitriy Rabotyagov","email":"noonedeadpunk@gmail.com","username":"noonedeadpunk"},"reviewer":{"_account_id":37598,"name":"Ivan Anfimov","display_name":"Ivan Anfimov","email":"lazekteam@gmail.com","username":"anfimovir"},"state":"CC"},{"updated":"2026-01-19 19:42:18.000000000","updated_by":{"_account_id":37598,"name":"Ivan Anfimov","display_name":"Ivan Anfimov","email":"lazekteam@gmail.com","username":"anfimovir"},"reviewer":{"_account_id":37598,"name":"Ivan Anfimov","display_name":"Ivan Anfimov","email":"lazekteam@gmail.com","username":"anfimovir"},"state":"REVIEWER"},{"updated":"2026-01-19 19:42:18.000000000","updated_by":{"_account_id":37598,"name":"Ivan Anfimov","display_name":"Ivan Anfimov","email":"lazekteam@gmail.com","username":"anfimovir"},"reviewer":{"_account_id":32666,"name":"Damian Dąbrowski","email":"damian@dabrowski.cloud","username":"ddabrowski"},"state":"REVIEWER"},{"updated":"2026-01-19 19:42:18.000000000","updated_by":{"_account_id":37598,"name":"Ivan Anfimov","display_name":"Ivan Anfimov","email":"lazekteam@gmail.com","username":"anfimovir"},"reviewer":{"_account_id":31542,"name":"Andrew Bonney","email":"andrew.bonney@bbc.co.uk","username":"andrewbonney"},"state":"REVIEWER"},{"updated":"2026-01-19 19:42:18.000000000","updated_by":{"_account_id":37598,"name":"Ivan Anfimov","display_name":"Ivan Anfimov","email":"lazekteam@gmail.com","username":"anfimovir"},"reviewer":{"_account_id":25023,"name":"Jonathan Rosser","email":"jonathan.rosser@rd.bbc.co.uk","username":"jrosser"},"state":"REVIEWER"},{"updated":"2026-01-19 20:38:14.000000000","updated_by":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"reviewer":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"state":"REVIEWER"}],"messages":[{"id":"bedb6751fbd415c69ccf82a03d9f7ec9681e68ca","tag":"autogenerated:gerrit:newPatchSet","author":{"_account_id":28619,"name":"Dmitriy Rabotyagov","email":"noonedeadpunk@gmail.com","username":"noonedeadpunk"},"date":"2026-01-19 18:43:11.000000000","message":"Uploaded patch set 1.","accounts_in_message":[],"_revision_number":1},{"id":"e1dd92b5788edc0b07a85831c8aaad45c8baf84a","author":{"_account_id":37598,"name":"Ivan Anfimov","display_name":"Ivan Anfimov","email":"lazekteam@gmail.com","username":"anfimovir"},"date":"2026-01-19 19:42:18.000000000","message":"Patch Set 1: Code-Review+1","accounts_in_message":[],"_revision_number":1},{"id":"509caedb7cc00eaff582d80050ddd1666b69093b","tag":"autogenerated:zuul:check","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2026-01-19 20:38:14.000000000","message":"Patch Set 1: Verified+1\n\nBuild succeeded (check pipeline).\nhttps://zuul.opendev.org/t/openstack/buildset/e49af7640a5248fb80b06b097fdfd380\n\n- openstack-ansible-linters-ubuntu-noble https://zuul.opendev.org/t/openstack/build/55b2532322a648d9a16b7bb8dd962eaf : SUCCESS in 10m 29s\n- openstack-tox-docs https://zuul.opendev.org/t/openstack/build/b4a6ea72fa4646298826ab3a499e2d04 : SUCCESS in 6m 54s\n- openstack-ansible-deploy-infra_lxc-centos-10-stream https://zuul.opendev.org/t/openstack/build/46372d677cd047778b1cad6b1711d90f : SUCCESS in 1h 05m 54s\n- openstack-ansible-deploy-infra_lxc-debian-bookworm https://zuul.opendev.org/t/openstack/build/2c683ccd1d544a3e8ac393175954cc54 : SUCCESS in 47m 34s\n- openstack-ansible-deploy-infra_lxc-debian-trixie https://zuul.opendev.org/t/openstack/build/0b8ed5e288354cd78a4d7132521fd7db : SUCCESS in 1h 16m 00s\n- openstack-ansible-deploy-infra_lxc-rockylinux-10 https://zuul.opendev.org/t/openstack/build/abcb2f639ff94cc38b47efff1d4d0c22 : SUCCESS in 1h 23m 38s\n- openstack-ansible-deploy-infra_lxc_tls-rockylinux-10 https://zuul.opendev.org/t/openstack/build/05c8cdb172c0431483c86ff584f0447c : SUCCESS in 49m 45s\n- openstack-ansible-deploy-infra_lxc-ubuntu-noble https://zuul.opendev.org/t/openstack/build/0120ae9626ad45a68a44d9235a0dbb60 : SUCCESS in 44m 07s\n- openstack-ansible-deploy-infra_lxc_stepca-ubuntu-noble https://zuul.opendev.org/t/openstack/build/fb9f27a9f8ba464faf353b4ab45b281d : SUCCESS in 47m 04s\n- openstack-ansible-deploy-infra_lxc_tls-ubuntu-noble https://zuul.opendev.org/t/openstack/build/afa5bc8b4fb34ef691bdcac6f670c2b1 : SUCCESS in 1h 22m 48s\n- openstack-ansible-upgrade-infra_lxc-ubuntu-noble https://zuul.opendev.org/t/openstack/build/0022b736f9d34e2d91020e7d7eda4e94 : SUCCESS in 1h 50m 42s\n- openstack-ansible-deploy-hosts_lxc_zfs-ubuntu-noble https://zuul.opendev.org/t/openstack/build/69c16788596d42ce89592ea4980fb5c5 : SUCCESS in 27m 48s\n- openstack-ansible-deploy-hosts_lxc_btrfs-ubuntu-noble https://zuul.opendev.org/t/openstack/build/c91294bdc88c43e0956f00f90923cc8b : SUCCESS in 31m 49s\n- openstack-ansible-deploy-hosts_lxc_overlayfs-ubuntu-noble https://zuul.opendev.org/t/openstack/build/cdfded2a7618440889e815df403898b7 : SUCCESS in 20m 38s\n- openstack-ansible-deploy-hosts_lxc_btrfs-debian-bookworm https://zuul.opendev.org/t/openstack/build/b97a1eb831114f7fa7f64708343d679e : SUCCESS in 26m 13s\n- openstack-ansible-deploy-hosts_lxc_overlayfs-debian-bookworm https://zuul.opendev.org/t/openstack/build/0ae14eb6153047d3bb64a9034194a94a : SUCCESS in 26m 03s\n- openstack-ansible-deploy-hosts_lxc_overlayfs-rockylinux-10 https://zuul.opendev.org/t/openstack/build/9a45542fb80c44c6902c6f9e97dc1cbe : SUCCESS in 23m 43s\n- openstack-ansible-deploy-hosts_lxc_btrfs-debian-trixie https://zuul.opendev.org/t/openstack/build/71dacef281e442e98c49dc00808825a7 : SUCCESS in 22m 34s\n- openstack-ansible-deploy-hosts_lxc_overlayfs-debian-trixie https://zuul.opendev.org/t/openstack/build/cb86d202ab09435f872d3058623880ca : SUCCESS in 25m 44s","accounts_in_message":[],"_revision_number":1},{"id":"cd914121340e6e6957a0614a22f55006ab3adbb2","author":{"_account_id":31542,"name":"Andrew Bonney","email":"andrew.bonney@bbc.co.uk","username":"andrewbonney"},"date":"2026-01-20 08:23:45.000000000","message":"Patch Set 1: Code-Review+2","accounts_in_message":[],"_revision_number":1},{"id":"c11e32132bced428746dffd32b8899ad0aa94e23","author":{"_account_id":28619,"name":"Dmitriy Rabotyagov","email":"noonedeadpunk@gmail.com","username":"noonedeadpunk"},"date":"2026-01-21 14:57:42.000000000","message":"Patch Set 1: Code-Review+2 Workflow+1","accounts_in_message":[],"_revision_number":1},{"id":"a9911ac0e5eb5e11a4fb543311496005c3d2e856","tag":"autogenerated:zuul:gate","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2026-01-21 14:57:45.000000000","message":"Patch Set 1: Verified-2\n\nBuild failed (gate pipeline).  For information on how to proceed, see\nhttps://docs.opendev.org/opendev/infra-manual/latest/developers.html#automated-testing\n\nhttps://zuul.opendev.org/t/openstack/buildset/37e9fffea5c84dabb0a4c2cf645a0546\n\n\nWarning:\n  Change 972873,2 in project openstack/openstack-ansible-rabbitmq_server does not share a change queue with 973887 in project openstack/openstack-ansible-lxc_hosts","accounts_in_message":[],"_revision_number":1},{"id":"faab56f74266f766bc9197139bbca729e86dcf61","author":{"_account_id":28619,"name":"Dmitriy Rabotyagov","email":"noonedeadpunk@gmail.com","username":"noonedeadpunk"},"date":"2026-01-21 16:18:00.000000000","message":"Patch Set 1:\n\n(1 comment)","accounts_in_message":[],"_revision_number":1},{"id":"42bde13afc1764bbbf2e758a40a593df0823e731","tag":"autogenerated:zuul:check","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2026-01-21 18:17:09.000000000","message":"Patch Set 1: Verified-1\n\nBuild failed (check pipeline).  For information on how to proceed, see\nhttps://docs.opendev.org/opendev/infra-manual/latest/developers.html#automated-testing\nand https://docs.openstack.org/project-team-guide/testing.html#how-to-handle-test-failures\n\nhttps://zuul.opendev.org/t/openstack/buildset/81e37cd9d0fb4dd894ef06b57f163ece\n\n- openstack-ansible-linters-ubuntu-noble https://zuul.opendev.org/t/openstack/build/055d2b733136468a877b7b70268ea76f : SUCCESS in 10m 27s\n- openstack-tox-docs https://zuul.opendev.org/t/openstack/build/b4a6602c6ffc406a8e32b6577ed678df : SUCCESS in 4m 06s\n- openstack-ansible-deploy-infra_lxc-centos-10-stream https://zuul.opendev.org/t/openstack/build/2e58b1e2fe6b44eba3a205efda8cb2ae : SUCCESS in 48m 14s\n- openstack-ansible-deploy-infra_lxc-debian-bookworm https://zuul.opendev.org/t/openstack/build/f3f364a2c9b54bb09bf9e29afd0c41d7 : SUCCESS in 42m 38s\n- openstack-ansible-deploy-infra_lxc-debian-trixie https://zuul.opendev.org/t/openstack/build/88927d65e9e0490591b994d1fa7c27f1 : SUCCESS in 1h 20m 47s\n- openstack-ansible-deploy-infra_lxc-rockylinux-10 https://zuul.opendev.org/t/openstack/build/ef6ee2764f1043ca99e926506f1f230c : SUCCESS in 46m 22s\n- openstack-ansible-deploy-infra_lxc_tls-rockylinux-10 https://zuul.opendev.org/t/openstack/build/2bc2f2289c9740ccb98201fe00cf388c : SUCCESS in 47m 59s\n- openstack-ansible-deploy-infra_lxc-ubuntu-noble https://zuul.opendev.org/t/openstack/build/7b9db91f0c0b4a978ececb7cf9de0e02 : SUCCESS in 1h 05m 26s\n- openstack-ansible-deploy-infra_lxc_stepca-ubuntu-noble https://zuul.opendev.org/t/openstack/build/3892fec41daf498babfffdbf12c6ec57 : SUCCESS in 1h 13m 58s\n- openstack-ansible-deploy-infra_lxc_tls-ubuntu-noble https://zuul.opendev.org/t/openstack/build/99e37c219acf4547bdd1b691827cc281 : SUCCESS in 1h 24m 57s\n- openstack-ansible-upgrade-infra_lxc-ubuntu-noble https://zuul.opendev.org/t/openstack/build/479de8bef0aa46ea88d1224490ea467d : SUCCESS in 1h 46m 09s\n- openstack-ansible-deploy-hosts_lxc_zfs-ubuntu-noble https://zuul.opendev.org/t/openstack/build/71f82817299b42bfa3b891c430ef6b03 : SUCCESS in 16m 45s\n- openstack-ansible-deploy-hosts_lxc_btrfs-ubuntu-noble https://zuul.opendev.org/t/openstack/build/5df352cc7c354a5d9799b00c86357e9f : SUCCESS in 30m 34s\n- openstack-ansible-deploy-hosts_lxc_overlayfs-ubuntu-noble https://zuul.opendev.org/t/openstack/build/ed9eded1320a42d18ceccad480ee4708 : SUCCESS in 27m 44s\n- openstack-ansible-deploy-hosts_lxc_btrfs-debian-bookworm https://zuul.opendev.org/t/openstack/build/f2dca66e590748ae940f20c5417afc4a : SUCCESS in 25m 59s\n- openstack-ansible-deploy-hosts_lxc_overlayfs-debian-bookworm https://zuul.opendev.org/t/openstack/build/189472c80b1b40cb9ef97d55d86b4357 : SUCCESS in 21m 52s\n- openstack-ansible-deploy-hosts_lxc_overlayfs-rockylinux-10 https://zuul.opendev.org/t/openstack/build/44a37f46f7c64a2c962c0f6dd8e27bb8 : SUCCESS in 16m 51s\n- openstack-ansible-deploy-hosts_lxc_btrfs-debian-trixie https://zuul.opendev.org/t/openstack/build/faf654144b5d4b55a7e64827e8027412 : FAILURE in 15m 58s\n- openstack-ansible-deploy-hosts_lxc_overlayfs-debian-trixie https://zuul.opendev.org/t/openstack/build/e96da80ad54f480d827db3ca79eb9c88 : SUCCESS in 25m 28s","accounts_in_message":[],"_revision_number":1},{"id":"168405f411540d06f807d0b98ef7acbf2c7c1b17","author":{"_account_id":37598,"name":"Ivan Anfimov","display_name":"Ivan Anfimov","email":"lazekteam@gmail.com","username":"anfimovir"},"date":"2026-01-21 20:25:45.000000000","message":"Patch Set 1:\n\n(1 comment)","accounts_in_message":[],"_revision_number":1},{"id":"581abe9c06bc6bec1fe212a76bc2429e6d5d36c2","tag":"autogenerated:zuul:check","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2026-01-21 22:01:55.000000000","message":"Patch Set 1: Verified+1\n\nBuild succeeded (check pipeline).\nhttps://zuul.opendev.org/t/openstack/buildset/76d7e53c63e74bb7ba157db36696bb5e\n\n- openstack-ansible-linters-ubuntu-noble https://zuul.opendev.org/t/openstack/build/2eb5ffa76a1d477aba4b9e4960132a52 : SUCCESS in 7m 25s\n- openstack-tox-docs https://zuul.opendev.org/t/openstack/build/d4373c00f2ce4956afd5e8babce8b7a7 : SUCCESS in 6m 35s\n- openstack-ansible-deploy-infra_lxc-centos-10-stream https://zuul.opendev.org/t/openstack/build/505970eba2e24ff88506182beda0cd0f : SUCCESS in 1h 00m 14s\n- openstack-ansible-deploy-infra_lxc-debian-bookworm https://zuul.opendev.org/t/openstack/build/4e8c619d1f604450a8a2cb45388ce01a : SUCCESS in 1h 20m 54s\n- openstack-ansible-deploy-infra_lxc-debian-trixie https://zuul.opendev.org/t/openstack/build/a856de32cff94d33864b4bfb89e06cf7 : SUCCESS in 1h 12m 18s\n- openstack-ansible-deploy-infra_lxc-rockylinux-10 https://zuul.opendev.org/t/openstack/build/1cac31dc66b34faaaed8a2b25da95a51 : SUCCESS in 1h 28m 33s\n- openstack-ansible-deploy-infra_lxc_tls-rockylinux-10 https://zuul.opendev.org/t/openstack/build/79f1299e5635436bb8174a4fd7f88055 : SUCCESS in 1h 04m 47s\n- openstack-ansible-deploy-infra_lxc-ubuntu-noble https://zuul.opendev.org/t/openstack/build/6471229f88044e3e9c547f5a9162d882 : SUCCESS in 1h 19m 45s\n- openstack-ansible-deploy-infra_lxc_stepca-ubuntu-noble https://zuul.opendev.org/t/openstack/build/16e9fca00f814ff28c92e81bab69bbbd : SUCCESS in 1h 19m 03s\n- openstack-ansible-deploy-infra_lxc_tls-ubuntu-noble https://zuul.opendev.org/t/openstack/build/fcc36f1687844a77946d6705d3618f3c : SUCCESS in 1h 18m 10s\n- openstack-ansible-upgrade-infra_lxc-ubuntu-noble https://zuul.opendev.org/t/openstack/build/3dc52d70618b4112a84ef86dea407045 : SUCCESS in 1h 20m 49s\n- openstack-ansible-deploy-hosts_lxc_zfs-ubuntu-noble https://zuul.opendev.org/t/openstack/build/1c6b438434c74b50b271a82c8325d630 : SUCCESS in 22m 30s\n- openstack-ansible-deploy-hosts_lxc_btrfs-ubuntu-noble https://zuul.opendev.org/t/openstack/build/e3bcf96cd8214fa1938f347c2c3c2335 : SUCCESS in 20m 32s\n- openstack-ansible-deploy-hosts_lxc_overlayfs-ubuntu-noble https://zuul.opendev.org/t/openstack/build/5d3a1520353c4f6296d82078e7ca8e29 : SUCCESS in 25m 24s\n- openstack-ansible-deploy-hosts_lxc_btrfs-debian-bookworm https://zuul.opendev.org/t/openstack/build/0a826b5c47ba4cb8a0a0b460e2779b7c : SUCCESS in 25m 45s\n- openstack-ansible-deploy-hosts_lxc_overlayfs-debian-bookworm https://zuul.opendev.org/t/openstack/build/be62f071b9b140989d55234650e66e15 : SUCCESS in 22m 27s\n- openstack-ansible-deploy-hosts_lxc_overlayfs-rockylinux-10 https://zuul.opendev.org/t/openstack/build/c135ecbc30c24d07825e2eabb4228711 : SUCCESS in 22m 39s\n- openstack-ansible-deploy-hosts_lxc_btrfs-debian-trixie https://zuul.opendev.org/t/openstack/build/f37184cecb6344b3ad8f850cd5b5bf1b : SUCCESS in 24m 12s\n- openstack-ansible-deploy-hosts_lxc_overlayfs-debian-trixie https://zuul.opendev.org/t/openstack/build/a68b85abe5df4361a066c95e0a9a484a : SUCCESS in 22m 54s","accounts_in_message":[],"_revision_number":1},{"id":"21f117b7f4e7845a84aafce38c0c62ebb001690f","tag":"autogenerated:zuul:gate","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2026-01-21 22:04:08.000000000","message":"Patch Set 1: -Verified\n\nStarting gate jobs.","accounts_in_message":[],"_revision_number":1},{"id":"692701b5b92c5e6196bb60be4ec62151fbd4dccc","tag":"autogenerated:zuul:gate","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2026-01-21 23:44:09.000000000","message":"Patch Set 1: Verified+2\n\nBuild succeeded (gate pipeline).\nhttps://zuul.opendev.org/t/openstack/buildset/f45d9ecca5174563b3562590ca324492\n\n- openstack-ansible-linters-ubuntu-noble https://zuul.opendev.org/t/openstack/build/988f17585bdd46808478e4097161476d : SUCCESS in 5m 25s\n- openstack-tox-docs https://zuul.opendev.org/t/openstack/build/a4efcc35cd9f4373ab4f95f20c597854 : SUCCESS in 5m 37s\n- openstack-ansible-deploy-infra_lxc-centos-10-stream https://zuul.opendev.org/t/openstack/build/c15f8aa148804f74a7e62878562c89f0 : SUCCESS in 1h 02m 04s\n- openstack-ansible-deploy-infra_lxc-debian-bookworm https://zuul.opendev.org/t/openstack/build/16bf1f809d1f42c7b8766a80ddd01325 : SUCCESS in 58m 26s\n- openstack-ansible-deploy-infra_lxc-debian-trixie https://zuul.opendev.org/t/openstack/build/fec13fe44af2458b8236a20650bb1b24 : SUCCESS in 1h 11m 46s\n- openstack-ansible-deploy-infra_lxc-rockylinux-10 https://zuul.opendev.org/t/openstack/build/51750fceca5043e19929541c3b700f22 : SUCCESS in 46m 42s\n- openstack-ansible-deploy-infra_lxc_tls-rockylinux-10 https://zuul.opendev.org/t/openstack/build/1f424ae9677e46e09016ef78b30ff794 : SUCCESS in 1h 28m 47s\n- openstack-ansible-deploy-infra_lxc-ubuntu-noble https://zuul.opendev.org/t/openstack/build/614e4f4b94c4442aa589ca12f7274b82 : SUCCESS in 1h 36m 40s\n- openstack-ansible-deploy-infra_lxc_stepca-ubuntu-noble https://zuul.opendev.org/t/openstack/build/3db57b7fbe2c4bee9361e21cec8b9647 : SUCCESS in 41m 32s\n- openstack-ansible-deploy-infra_lxc_tls-ubuntu-noble https://zuul.opendev.org/t/openstack/build/49d6042609c74ba7a2769158b87f5a2b : SUCCESS in 1h 23m 34s\n- openstack-ansible-upgrade-infra_lxc-ubuntu-noble https://zuul.opendev.org/t/openstack/build/8bd90923b43f4deea4127d23aaf6c90d : SUCCESS in 1h 29m 36s\n- openstack-ansible-deploy-hosts_lxc_zfs-ubuntu-noble https://zuul.opendev.org/t/openstack/build/0e9a2e1b1331442388635d2103197427 : SUCCESS in 26m 55s\n- openstack-ansible-deploy-hosts_lxc_btrfs-ubuntu-noble https://zuul.opendev.org/t/openstack/build/f209aff97f0e4907af4fbddbec42b6a7 : SUCCESS in 14m 31s\n- openstack-ansible-deploy-hosts_lxc_overlayfs-ubuntu-noble https://zuul.opendev.org/t/openstack/build/5b7f0cdb66fd484d954a4890eae7ba31 : SUCCESS in 24m 29s\n- openstack-ansible-deploy-hosts_lxc_btrfs-debian-bookworm https://zuul.opendev.org/t/openstack/build/cd542b5e888941bfa2ffd0a97714bfd1 : SUCCESS in 14m 46s\n- openstack-ansible-deploy-hosts_lxc_overlayfs-debian-bookworm https://zuul.opendev.org/t/openstack/build/86330de156c94774b21a22ef75fffbdf : SUCCESS in 26m 09s\n- openstack-ansible-deploy-hosts_lxc_btrfs-debian-trixie https://zuul.opendev.org/t/openstack/build/4270d541c3c84088af3e223db5413124 : SUCCESS in 21m 16s\n- openstack-ansible-deploy-hosts_lxc_overlayfs-debian-trixie https://zuul.opendev.org/t/openstack/build/18e98d7ca31d4f55b2abce277032da72 : SUCCESS in 19m 45s\n- openstack-ansible-deploy-hosts_lxc_overlayfs-rockylinux-10 https://zuul.opendev.org/t/openstack/build/5e4cd545411c423e891b34c9f9a3cf04 : SUCCESS in 21m 58s","accounts_in_message":[],"_revision_number":1},{"id":"29efbe6bb4518524388deb8958bdbfd0571cbd39","tag":"autogenerated:gerrit:merged","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2026-01-21 23:44:09.000000000","message":"Change has been successfully merged","accounts_in_message":[],"_revision_number":1},{"id":"cc31553a5046fadba2a1a14e9fa578d6831e51a9","tag":"autogenerated:zuul:promote","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2026-01-21 23:45:33.000000000","message":"Patch Set 1:\n\nBuild succeeded (promote pipeline).\nhttps://zuul.opendev.org/t/openstack/buildset/c473a7b2d42d437589093435a0a17ca1\n\n- promote-openstack-tox-docs https://zuul.opendev.org/t/openstack/build/d3d096cf01454fe5911855d780dc33a8 : SUCCESS in 36s","accounts_in_message":[],"_revision_number":1}],"current_revision_number":1,"current_revision":"f549598cf0d3f5977ae5f0413ff0266b0d953eee","revisions":{"f549598cf0d3f5977ae5f0413ff0266b0d953eee":{"kind":"REWORK","_number":1,"created":"2026-01-19 18:43:11.000000000","uploader":{"_account_id":28619,"name":"Dmitriy Rabotyagov","email":"noonedeadpunk@gmail.com","username":"noonedeadpunk"},"ref":"refs/changes/87/973887/1","fetch":{"anonymous http":{"url":"https://review.opendev.org/openstack/openstack-ansible-lxc_hosts","ref":"refs/changes/87/973887/1","commands":{"Checkout":"git fetch https://review.opendev.org/openstack/openstack-ansible-lxc_hosts refs/changes/87/973887/1 \u0026\u0026 git checkout FETCH_HEAD","Cherry Pick":"git fetch https://review.opendev.org/openstack/openstack-ansible-lxc_hosts refs/changes/87/973887/1 \u0026\u0026 git cherry-pick FETCH_HEAD","Format Patch":"git fetch https://review.opendev.org/openstack/openstack-ansible-lxc_hosts refs/changes/87/973887/1 \u0026\u0026 git format-patch -1 --stdout FETCH_HEAD","Pull":"git pull https://review.opendev.org/openstack/openstack-ansible-lxc_hosts refs/changes/87/973887/1"}}},"commit":{"parents":[{"commit":"ffec6ff38ff4aee5663d227ac55e180421906de9","subject":"Do not install post-quantum GPG keys","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/openstack-ansible-lxc_hosts/commit/ffec6ff38ff4aee5663d227ac55e180421906de9"}]}],"author":{"name":"Ivan Anfimov","email":"lazekteam@gmail.com","date":"2026-01-08 20:42:18.000000000","tz":0},"committer":{"name":"Dmitriy Rabotyagov","email":"dmitriy.rabotyagov@cleura.com","date":"2026-01-19 18:42:43.000000000","tz":60},"subject":"Revert \"Temporary disable RHEL jobs\"","message":"Revert \"Temporary disable RHEL jobs\"\n\nThis reverts commit b213e5012bfe538971cf77889c12128a5b9e3592.\n\nDepends-On: https://review.opendev.org/c/openstack/openstack-ansible-rabbitmq_server/+/972873\nChange-Id: Ib1c557950c45276b26a8584bcae4891fed2aaa71\nSigned-off-by: Dmitriy Rabotyagov \u003cdmitriy.rabotyagov@cleura.com\u003e\n","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/openstack-ansible-lxc_hosts/commit/f549598cf0d3f5977ae5f0413ff0266b0d953eee"}],"resolve_conflicts_web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/openstack-ansible-lxc_hosts/commit/f549598cf0d3f5977ae5f0413ff0266b0d953eee"}]},"branch":"refs/heads/stable/2025.2"}},"requirements":[],"submit_records":[{"rule_name":"gerrit~DefaultSubmitRule","status":"CLOSED","labels":[{"label":"Verified","status":"MAY","applied_by":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]}},{"label":"Code-Review","status":"MAY","applied_by":{"_account_id":28619,"name":"Dmitriy Rabotyagov","email":"noonedeadpunk@gmail.com","username":"noonedeadpunk"}},{"label":"Workflow","status":"MAY","applied_by":{"_account_id":28619,"name":"Dmitriy Rabotyagov","email":"noonedeadpunk@gmail.com","username":"noonedeadpunk"}},{"label":"Backport-Candidate","status":"MAY"}]}],"submit_requirements":[{"name":"Verified","description":"Verified in gate by CI","status":"SATISFIED","is_legacy":false,"submittability_expression_result":{"expression":"label:Verified\u003dMAX AND -label:Verified\u003dMIN","fulfilled":true,"status":"PASS","passing_atoms":["label:Verified\u003dMAX"],"failing_atoms":["label:Verified\u003dMIN"],"atom_explanations":{}}},{"name":"Backport-Candidate","description":"Backport candidate status","status":"NOT_APPLICABLE","is_legacy":false,"applicability_expression_result":{"fulfilled":false,"status":"FAIL"},"submittability_expression_result":{"expression":"is:true","fulfilled":true,"status":"NOT_EVALUATED","passing_atoms":[],"failing_atoms":[],"atom_explanations":{}}},{"name":"Code-Review","description":"Code reviewed by core reviewer","status":"SATISFIED","is_legacy":false,"submittability_expression_result":{"expression":"label:Code-Review\u003dMAX AND -label:Code-Review\u003dMIN","fulfilled":true,"status":"PASS","passing_atoms":["label:Code-Review\u003dMAX"],"failing_atoms":["label:Code-Review\u003dMIN"],"atom_explanations":{}}},{"name":"Workflow","description":"Approved for gate by core reviewer","status":"SATISFIED","is_legacy":false,"submittability_expression_result":{"expression":"label:Workflow\u003dMAX AND -label:Workflow\u003dMIN","fulfilled":true,"status":"PASS","passing_atoms":["label:Workflow\u003dMAX"],"failing_atoms":["label:Workflow\u003dMIN"],"atom_explanations":{}}}]}
