)]}'
{"id":"openstack%2Fopenstack-ansible-lxc_hosts~985216","triplet_id":"openstack%2Fopenstack-ansible-lxc_hosts~master~I2ceb1040e978dfe4fcd552dfd67d68050f552b3c","project":"openstack/openstack-ansible-lxc_hosts","branch":"master","topic":"osa/resolute","attention_set":{},"removed_from_attention_set":{"28619":{"account":{"_account_id":28619,"name":"Dmitriy Rabotyagov","email":"noonedeadpunk@gmail.com","username":"noonedeadpunk"},"last_update":"2026-05-25 10:00:22.000000000","reason":"\u003cGERRIT_ACCOUNT_28619\u003e replied on the change","reason_account":{"_account_id":28619,"name":"Dmitriy Rabotyagov","email":"noonedeadpunk@gmail.com","username":"noonedeadpunk"}}},"hashtags":[],"change_id":"I2ceb1040e978dfe4fcd552dfd67d68050f552b3c","subject":"Add support for Ubuntu 26.04","status":"MERGED","created":"2026-04-18 09:10:32.000000000","updated":"2026-05-25 14:51:18.000000000","submitted":"2026-05-25 14:50:16.000000000","submitter":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"total_comment_count":10,"unresolved_comment_count":0,"has_review_started":true,"submission_id":"985216-osa/resolute","meta_rev_id":"2ad26c52f03ace535540283037a03ac62e1e4e59","_number":985216,"virtual_id_number":985216,"owner":{"_account_id":28619,"name":"Dmitriy Rabotyagov","email":"noonedeadpunk@gmail.com","username":"noonedeadpunk"},"actions":{},"labels":{"Verified":{"approved":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"all":[{"value":0,"_account_id":32666,"name":"Damian Dąbrowski","email":"damian@dabrowski.cloud","username":"ddabrowski"},{"tag":"autogenerated:zuul:gate","value":2,"date":"2026-05-25 14:50:16.000000000","permitted_voting_range":{"min":2,"max":2},"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},{"value":0,"_account_id":25023,"name":"Jonathan Rosser","email":"jonathan.rosser@rd.bbc.co.uk","username":"jrosser"}],"values":{"-2":"Fails","-1":"Doesn\u0027t seem to work"," 0":"No score","+1":"Works for me","+2":"Verified"},"description":"","default_value":0,"optional":true},"Code-Review":{"approved":{"_account_id":32666,"name":"Damian Dąbrowski","email":"damian@dabrowski.cloud","username":"ddabrowski"},"all":[{"value":2,"date":"2026-05-14 11:27:26.000000000","permitted_voting_range":{"min":2,"max":2},"_account_id":32666,"name":"Damian Dąbrowski","email":"damian@dabrowski.cloud","username":"ddabrowski"},{"value":0,"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},{"value":2,"date":"2026-05-13 08:38:54.000000000","permitted_voting_range":{"min":2,"max":2},"_account_id":25023,"name":"Jonathan Rosser","email":"jonathan.rosser@rd.bbc.co.uk","username":"jrosser"}],"values":{"-2":"Do not merge","-1":"This patch needs further work before it can be merged"," 0":"No score","+1":"Looks good to me, but someone else must approve","+2":"Looks good to me (core reviewer)"},"description":"","default_value":0,"optional":true},"Workflow":{"approved":{"_account_id":32666,"name":"Damian Dąbrowski","email":"damian@dabrowski.cloud","username":"ddabrowski"},"all":[{"value":1,"date":"2026-05-14 11:27:26.000000000","permitted_voting_range":{"min":1,"max":1},"_account_id":32666,"name":"Damian Dąbrowski","email":"damian@dabrowski.cloud","username":"ddabrowski"},{"value":0,"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},{"value":0,"_account_id":25023,"name":"Jonathan Rosser","email":"jonathan.rosser@rd.bbc.co.uk","username":"jrosser"}],"values":{"-1":"Work in progress"," 0":"Ready for reviews","+1":"Approved"},"description":"","default_value":0,"optional":true},"Backport-Candidate":{"all":[{"value":0,"_account_id":32666,"name":"Damian Dąbrowski","email":"damian@dabrowski.cloud","username":"ddabrowski"},{"value":0,"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},{"value":0,"_account_id":25023,"name":"Jonathan Rosser","email":"jonathan.rosser@rd.bbc.co.uk","username":"jrosser"}],"values":{"-1":"Not A Backport Candidate"," 0":"Backport Review Needed","+1":"Proposed Backport"},"description":"","default_value":0,"optional":true}},"removable_reviewers":[],"reviewers":{"REVIEWER":[{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},{"_account_id":25023,"name":"Jonathan Rosser","email":"jonathan.rosser@rd.bbc.co.uk","username":"jrosser"},{"_account_id":32666,"name":"Damian Dąbrowski","email":"damian@dabrowski.cloud","username":"ddabrowski"}],"CC":[{"_account_id":37598,"name":"Ivan Anfimov","display_name":"Ivan Anfimov","email":"lazekteam@gmail.com","username":"anfimovir"}]},"pending_reviewers":{},"reviewer_updates":[{"updated":"2026-04-18 12:03:59.000000000","updated_by":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"reviewer":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"state":"REVIEWER"},{"updated":"2026-04-20 10:06:28.000000000","updated_by":{"_account_id":37598,"name":"Ivan Anfimov","display_name":"Ivan Anfimov","email":"lazekteam@gmail.com","username":"anfimovir"},"reviewer":{"_account_id":37598,"name":"Ivan Anfimov","display_name":"Ivan Anfimov","email":"lazekteam@gmail.com","username":"anfimovir"},"state":"CC"},{"updated":"2026-05-13 08:38:54.000000000","updated_by":{"_account_id":25023,"name":"Jonathan Rosser","email":"jonathan.rosser@rd.bbc.co.uk","username":"jrosser"},"reviewer":{"_account_id":25023,"name":"Jonathan Rosser","email":"jonathan.rosser@rd.bbc.co.uk","username":"jrosser"},"state":"REVIEWER"},{"updated":"2026-05-14 11:27:26.000000000","updated_by":{"_account_id":32666,"name":"Damian Dąbrowski","email":"damian@dabrowski.cloud","username":"ddabrowski"},"reviewer":{"_account_id":32666,"name":"Damian Dąbrowski","email":"damian@dabrowski.cloud","username":"ddabrowski"},"state":"REVIEWER"}],"messages":[{"id":"f0c7b54cf05c75c168c23eeb1c0d2159eda2700a","tag":"autogenerated:gerrit:newPatchSet","author":{"_account_id":28619,"name":"Dmitriy Rabotyagov","email":"noonedeadpunk@gmail.com","username":"noonedeadpunk"},"date":"2026-04-18 09:10:32.000000000","message":"Uploaded patch set 1.","accounts_in_message":[],"_revision_number":1},{"id":"ab55ef4587d4cb3f24294f7c132869fc4cdf44fb","tag":"autogenerated:gerrit:newPatchSet","author":{"_account_id":28619,"name":"Dmitriy Rabotyagov","email":"noonedeadpunk@gmail.com","username":"noonedeadpunk"},"date":"2026-04-18 09:45:58.000000000","message":"Uploaded patch set 2.","accounts_in_message":[],"_revision_number":2},{"id":"2bdad0125b5de082e4572a5f62fcf49130e5b161","tag":"autogenerated:gerrit:newPatchSet","author":{"_account_id":28619,"name":"Dmitriy Rabotyagov","email":"noonedeadpunk@gmail.com","username":"noonedeadpunk"},"date":"2026-04-18 09:57:51.000000000","message":"Uploaded patch set 3.","accounts_in_message":[],"_revision_number":3},{"id":"546d2a3636ed33e37a96a56cf6ef613408b3c256","tag":"autogenerated:zuul:check","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2026-04-18 12:03:59.000000000","message":"Patch Set 3: Verified-1\n\nBuild failed (check pipeline).  For information on how to proceed, see\nhttps://docs.opendev.org/opendev/infra-manual/latest/developers.html#automated-testing\nand https://docs.openstack.org/project-team-guide/testing.html#how-to-handle-test-failures\n\nhttps://zuul.opendev.org/t/openstack/buildset/e5e0c34326f84dd5bd46861896141761\n\n- openstack-ansible-linters-ubuntu-noble https://zuul.opendev.org/t/openstack/build/5363c9b3047c4ff6b0f1ce02032be1d6 : POST_FAILURE in 11m 57s\n- openstack-tox-docs https://zuul.opendev.org/t/openstack/build/38120a1dbc344caa8f2f9e5946eb80f8 : POST_FAILURE in 6m 45s\n- openstack-ansible-deploy-infra_lxc-centos-10-stream https://zuul.opendev.org/t/openstack/build/166883abdbff4a9e8a155fd34e35f651 : POST_FAILURE in 1h 30m 09s\n- openstack-ansible-deploy-infra_lxc-debian-bookworm https://zuul.opendev.org/t/openstack/build/0390ba6fbb524af38c4a718e1e5bb8d4 : POST_FAILURE in 14m 41s\n- openstack-ansible-deploy-infra_lxc-debian-trixie https://zuul.opendev.org/t/openstack/build/2c6c4df7c0fc441bad446c45156a8c79 : POST_FAILURE in 1h 10m 02s\n- openstack-ansible-deploy-infra_lxc-rockylinux-10 https://zuul.opendev.org/t/openstack/build/ca7a46f0f59c4bb18a1014ff1c82bf16 : POST_FAILURE in 1h 56m 11s\n- openstack-ansible-deploy-infra_lxc_tls-rockylinux-10 https://zuul.opendev.org/t/openstack/build/7f9bea9ca03b4fa19b3a31785e857ed1 : POST_FAILURE in 1h 58m 17s\n- openstack-ansible-deploy-infra_lxc-ubuntu-noble https://zuul.opendev.org/t/openstack/build/5e755a1203ac462c8a82df2565508afb : POST_FAILURE in 1h 28m 17s\n- openstack-ansible-deploy-infra_lxc_stepca-ubuntu-noble https://zuul.opendev.org/t/openstack/build/6a928c98566a4aeeaede645adc3dbbb9 : POST_FAILURE in 1h 28m 00s\n- openstack-ansible-deploy-infra_lxc_tls-ubuntu-noble https://zuul.opendev.org/t/openstack/build/c4ca71bf5eda4147b469c782712495f1 : POST_FAILURE in 1h 28m 00s\n- openstack-ansible-upgrade-infra_lxc-ubuntu-noble https://zuul.opendev.org/t/openstack/build/d4e1532fea1e439f80fb1545b7c84db5 : POST_FAILURE in 1h 54m 09s\n- openstack-ansible-deploy-hosts_lxc_zfs-ubuntu-noble https://zuul.opendev.org/t/openstack/build/82182091df4f4303b3af242f58fc53a8 : POST_FAILURE in 30m 08s\n- openstack-ansible-deploy-hosts_lxc_btrfs-ubuntu-noble https://zuul.opendev.org/t/openstack/build/aa3dba66f10247ba88db66caa91b442e : POST_FAILURE in 31m 28s\n- openstack-ansible-deploy-hosts_lxc_overlayfs-ubuntu-noble https://zuul.opendev.org/t/openstack/build/2fc2d93ef34c46bb975b5d13403854a4 : POST_FAILURE in 31m 19s\n- openstack-ansible-deploy-hosts_lxc_btrfs-debian-bookworm https://zuul.opendev.org/t/openstack/build/943b2530df594aa7a214c6d0be06b5ce : POST_FAILURE in 19m 55s\n- openstack-ansible-deploy-hosts_lxc_overlayfs-debian-bookworm https://zuul.opendev.org/t/openstack/build/dd39ae468b694b50af69b80996f98c49 : POST_FAILURE in 19m 57s\n- openstack-ansible-deploy-hosts_lxc_overlayfs-rockylinux-10 https://zuul.opendev.org/t/openstack/build/0356c60376b34091affea2179c2ab695 : POST_FAILURE in 26m 59s\n- openstack-ansible-deploy-hosts_lxc_btrfs-debian-trixie https://zuul.opendev.org/t/openstack/build/bc3cae490c3a49d1b4848f34099c9a90 : POST_FAILURE in 29m 20s\n- openstack-ansible-deploy-hosts_lxc_overlayfs-debian-trixie https://zuul.opendev.org/t/openstack/build/2dcaaf48c9314f4295269f2b110ae17f : POST_FAILURE in 28m 58s","accounts_in_message":[],"_revision_number":3},{"id":"f45038cab2476e862684924f26a132181b99ddbe","author":{"_account_id":37598,"name":"Ivan Anfimov","display_name":"Ivan Anfimov","email":"lazekteam@gmail.com","username":"anfimovir"},"date":"2026-04-20 10:06:28.000000000","message":"Patch Set 3:\n\n(1 comment)","accounts_in_message":[],"_revision_number":3},{"id":"e4d22cf051ad0b4bd5e6f8a71ebe1b30de4fb5ed","tag":"autogenerated:zuul:check","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2026-04-20 13:35:46.000000000","message":"Patch Set 3:\n\nBuild failed (check pipeline).  For information on how to proceed, see\nhttps://docs.opendev.org/opendev/infra-manual/latest/developers.html#automated-testing\nand https://docs.openstack.org/project-team-guide/testing.html#how-to-handle-test-failures\n\nhttps://zuul.opendev.org/t/openstack/buildset/feb8284541f24ef3afb90db46790ffff\n\n- openstack-ansible-linters-ubuntu-noble https://zuul.opendev.org/t/openstack/build/0dd7be16e50a4ccb98e7e7c44c45e007 : SUCCESS in 14m 28s\n- openstack-tox-docs https://zuul.opendev.org/t/openstack/build/63a31e986d454b41a3a6f5d89c3d067d : SUCCESS in 6m 45s\n- openstack-ansible-deploy-infra_lxc-centos-10-stream https://zuul.opendev.org/t/openstack/build/bd3a09167a0349dfa337f9f023630897 : SUCCESS in 1h 32m 58s\n- openstack-ansible-deploy-infra_lxc-debian-bookworm https://zuul.opendev.org/t/openstack/build/ef6a5edbed574359869aafe4b74b3248 : FAILURE in 16m 32s\n- openstack-ansible-deploy-infra_lxc-debian-trixie https://zuul.opendev.org/t/openstack/build/ec30217d58ff4bae93c99ed513ba6bc4 : SUCCESS in 1h 23m 15s\n- openstack-ansible-deploy-infra_lxc-rockylinux-10 https://zuul.opendev.org/t/openstack/build/1e9202df7a4948369fc5e2006ca8adaf : SUCCESS in 1h 38m 03s\n- openstack-ansible-deploy-infra_lxc_tls-rockylinux-10 https://zuul.opendev.org/t/openstack/build/80721eaee14142958d2536fe55fbbf66 : SUCCESS in 1h 40m 33s\n- openstack-ansible-deploy-infra_lxc-ubuntu-noble https://zuul.opendev.org/t/openstack/build/585836c3576149299224258ce5e34964 : SUCCESS in 1h 24m 14s\n- openstack-ansible-deploy-infra_lxc_stepca-ubuntu-noble https://zuul.opendev.org/t/openstack/build/c88f4f848e394c89901f40cd3f1a38fb : SUCCESS in 2h 22m 40s\n- openstack-ansible-deploy-infra_lxc_tls-ubuntu-noble https://zuul.opendev.org/t/openstack/build/d7f3e3b487f84e778ac65f09e2af4c5f : TIMED_OUT in 3h 06m 27s\n- openstack-ansible-upgrade-infra_lxc-ubuntu-noble https://zuul.opendev.org/t/openstack/build/091e3b8a67bb4a008e314c129b0bf27e : SUCCESS in 1h 58m 15s\n- openstack-ansible-deploy-hosts_lxc_zfs-ubuntu-noble https://zuul.opendev.org/t/openstack/build/d88b8998772244cc8bf2fdf6843d0df5 : SUCCESS in 33m 34s\n- openstack-ansible-deploy-hosts_lxc_btrfs-ubuntu-noble https://zuul.opendev.org/t/openstack/build/5e104c5473c54d38a9959320f10a6da7 : SUCCESS in 32m 05s\n- openstack-ansible-deploy-hosts_lxc_overlayfs-ubuntu-noble https://zuul.opendev.org/t/openstack/build/190cb280f5604c859f4df2ccc44100b6 : SUCCESS in 1h 08m 52s\n- openstack-ansible-deploy-hosts_lxc_btrfs-debian-bookworm https://zuul.opendev.org/t/openstack/build/5c2df2c4e76b4b5bb4aba179d7dc86fc : FAILURE in 22m 35s\n- openstack-ansible-deploy-hosts_lxc_overlayfs-debian-bookworm https://zuul.opendev.org/t/openstack/build/a78365ba9a0b4434af9748585b088d9c : FAILURE in 25m 37s\n- openstack-ansible-deploy-hosts_lxc_overlayfs-rockylinux-10 https://zuul.opendev.org/t/openstack/build/519190b8349548f68c71ede7179e5d42 : SUCCESS in 22m 11s\n- openstack-ansible-deploy-hosts_lxc_btrfs-debian-trixie https://zuul.opendev.org/t/openstack/build/d584626edaa240799c19973e38a05241 : SUCCESS in 23m 36s\n- openstack-ansible-deploy-hosts_lxc_overlayfs-debian-trixie https://zuul.opendev.org/t/openstack/build/50bfc2d6846d440fa9e6cf065b405396 : SUCCESS in 31m 10s","accounts_in_message":[],"_revision_number":3},{"id":"cb777996bab221bb7a8264e1fddddec2cc0d6897","author":{"_account_id":28619,"name":"Dmitriy Rabotyagov","email":"noonedeadpunk@gmail.com","username":"noonedeadpunk"},"date":"2026-04-20 17:51:03.000000000","message":"Patch Set 3:\n\n(1 comment)","accounts_in_message":[],"_revision_number":3},{"id":"47fd2dc14f5e8b0237f6bccb05730cd66195cb54","tag":"autogenerated:zuul:check","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2026-04-20 20:09:46.000000000","message":"Patch Set 3:\n\nBuild failed (check pipeline).  For information on how to proceed, see\nhttps://docs.opendev.org/opendev/infra-manual/latest/developers.html#automated-testing\nand https://docs.openstack.org/project-team-guide/testing.html#how-to-handle-test-failures\n\nhttps://zuul.opendev.org/t/openstack/buildset/59dabbb63b7342d3b14e928256076be5\n\n- openstack-ansible-linters-ubuntu-noble https://zuul.opendev.org/t/openstack/build/2c31cc9bc78747b28b54d5826e30d945 : SUCCESS in 14m 32s\n- openstack-tox-docs https://zuul.opendev.org/t/openstack/build/f206a88cc64248f294515824a8c887cc : SUCCESS in 6m 16s\n- openstack-ansible-deploy-infra_lxc-centos-10-stream https://zuul.opendev.org/t/openstack/build/3a323d12f94443ea9d5731fefcaf739b : SUCCESS in 1h 09m 29s\n- openstack-ansible-deploy-infra_lxc-debian-bookworm https://zuul.opendev.org/t/openstack/build/058a6e30cdf643da9122b65fb636b94f : FAILURE in 18m 33s\n- openstack-ansible-deploy-infra_lxc-debian-trixie https://zuul.opendev.org/t/openstack/build/1543f0e474874b588d5ae03c779c2d58 : SUCCESS in 51m 35s\n- openstack-ansible-deploy-infra_lxc-rockylinux-10 https://zuul.opendev.org/t/openstack/build/53d4a571303a447296cfab4cafb28491 : SUCCESS in 1h 21m 53s\n- openstack-ansible-deploy-infra_lxc_tls-rockylinux-10 https://zuul.opendev.org/t/openstack/build/c324261db39a46c1ae1ca57f78ebf9fd : SUCCESS in 55m 10s\n- openstack-ansible-deploy-infra_lxc-ubuntu-noble https://zuul.opendev.org/t/openstack/build/bd6cb63e187540f489070517d306c611 : SUCCESS in 1h 34m 50s\n- openstack-ansible-deploy-infra_lxc_stepca-ubuntu-noble https://zuul.opendev.org/t/openstack/build/9744e37ae16149f090c2ef62e3e67950 : SUCCESS in 1h 40m 32s\n- openstack-ansible-deploy-infra_lxc_tls-ubuntu-noble https://zuul.opendev.org/t/openstack/build/2238f8c424194227ae03c870ccf5b793 : SUCCESS in 1h 11m 26s\n- openstack-ansible-upgrade-infra_lxc-ubuntu-noble https://zuul.opendev.org/t/openstack/build/6abc0ab02a0b4b8292f1591bcd02b9d2 : SUCCESS in 1h 13m 00s\n- openstack-ansible-deploy-hosts_lxc_zfs-ubuntu-noble https://zuul.opendev.org/t/openstack/build/f0c5178e1af449e7ab9adc9b9d8fc70c : SUCCESS in 32m 43s\n- openstack-ansible-deploy-hosts_lxc_btrfs-ubuntu-noble https://zuul.opendev.org/t/openstack/build/8a26f1f2d127443a8d025d30c42446ae : SUCCESS in 22m 12s\n- openstack-ansible-deploy-hosts_lxc_overlayfs-ubuntu-noble https://zuul.opendev.org/t/openstack/build/36c765900e4049e39108239136965ec2 : SUCCESS in 28m 16s\n- openstack-ansible-deploy-hosts_lxc_btrfs-debian-bookworm https://zuul.opendev.org/t/openstack/build/9f27c7ca2fd64526abab7353bb663ce8 : FAILURE in 27m 05s\n- openstack-ansible-deploy-hosts_lxc_overlayfs-debian-bookworm https://zuul.opendev.org/t/openstack/build/6036d27ec82d43bbb8e7fa8d95dd9fc7 : FAILURE in 17m 12s\n- openstack-ansible-deploy-hosts_lxc_overlayfs-rockylinux-10 https://zuul.opendev.org/t/openstack/build/7a0b4f6fac8e48e1ae9dddbc26c21bcb : SUCCESS in 28m 28s\n- openstack-ansible-deploy-hosts_lxc_btrfs-debian-trixie https://zuul.opendev.org/t/openstack/build/08934eb37cc14dd5ae6197172864d5ec : SUCCESS in 25m 25s\n- openstack-ansible-deploy-hosts_lxc_overlayfs-debian-trixie https://zuul.opendev.org/t/openstack/build/3788a22bfe314076a1316d5ca059c8f5 : SUCCESS in 30m 38s","accounts_in_message":[],"_revision_number":3},{"id":"ca8d2185f1fed458ed2eca4105fb5ec0066b7815","tag":"autogenerated:gerrit:newPatchSet","author":{"_account_id":28619,"name":"Dmitriy Rabotyagov","email":"noonedeadpunk@gmail.com","username":"noonedeadpunk"},"date":"2026-04-21 08:35:22.000000000","message":"Patch Set 4: Published edit on patch set 3.\n\nOutdated Votes:\n* Verified-1 (copy condition: \"NEVER\")\n","accounts_in_message":[],"_revision_number":4},{"id":"ea664621e35946622a345b62ded56706af38771a","tag":"autogenerated:zuul:check","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2026-04-21 12:03:52.000000000","message":"Patch Set 4: Verified+1\n\nBuild succeeded (check pipeline).\nhttps://zuul.opendev.org/t/openstack/buildset/c76c268714b5417f932377469e482a45\n\n- openstack-ansible-linters-ubuntu-noble https://zuul.opendev.org/t/openstack/build/a19bf6287b284dd28e7633df04113d3b : SUCCESS in 15m 02s\n- openstack-tox-docs https://zuul.opendev.org/t/openstack/build/cb773f72cd174331a69e9aafee3071bd : SUCCESS in 6m 37s\n- openstack-ansible-deploy-infra_lxc-centos-10-stream https://zuul.opendev.org/t/openstack/build/3eee31a2eb40449c99dfac9cc291ccd3 : SUCCESS in 1h 17m 28s\n- openstack-ansible-deploy-infra_lxc-debian-bookworm https://zuul.opendev.org/t/openstack/build/0b123e13412e483fa37d571e66633618 : SUCCESS in 1h 22m 26s\n- openstack-ansible-deploy-infra_lxc-debian-trixie https://zuul.opendev.org/t/openstack/build/e913e3fd557d42859d3e02ce466e033f : SUCCESS in 1h 18m 09s\n- openstack-ansible-deploy-infra_lxc-rockylinux-10 https://zuul.opendev.org/t/openstack/build/935d841998104b7c8223b0f173049a82 : SUCCESS in 56m 01s\n- openstack-ansible-deploy-infra_lxc_tls-rockylinux-10 https://zuul.opendev.org/t/openstack/build/4b9648940144407586d2446da6e8847f : SUCCESS in 1h 22m 50s\n- openstack-ansible-deploy-infra_lxc-ubuntu-noble https://zuul.opendev.org/t/openstack/build/c3458eb628f942f1a649aef417fd1ef1 : SUCCESS in 1h 56m 36s\n- openstack-ansible-deploy-infra_lxc_stepca-ubuntu-noble https://zuul.opendev.org/t/openstack/build/6db33f2a55b149ffbec732e23cacf7d0 : SUCCESS in 1h 33m 00s\n- openstack-ansible-deploy-infra_lxc_tls-ubuntu-noble https://zuul.opendev.org/t/openstack/build/e6a17fb771974a46bc2a2f6041caac60 : SUCCESS in 1h 29m 24s\n- openstack-ansible-upgrade-infra_lxc-ubuntu-noble https://zuul.opendev.org/t/openstack/build/91b6b685eb0546cd85e425cde8407936 : SUCCESS in 2h 09m 59s\n- openstack-ansible-deploy-hosts_lxc_zfs-ubuntu-noble https://zuul.opendev.org/t/openstack/build/d4a924ce96964ca487ec53a662e99efe : SUCCESS in 36m 28s\n- openstack-ansible-deploy-hosts_lxc_btrfs-ubuntu-noble https://zuul.opendev.org/t/openstack/build/71f7105a5f144f96b2293c5d99087e2c : SUCCESS in 44m 18s\n- openstack-ansible-deploy-hosts_lxc_overlayfs-ubuntu-noble https://zuul.opendev.org/t/openstack/build/aa463449f17a43458475311f18424e93 : SUCCESS in 45m 50s\n- openstack-ansible-deploy-hosts_lxc_btrfs-debian-bookworm https://zuul.opendev.org/t/openstack/build/5556caa6649b4c9f83bafa821cb9af44 : SUCCESS in 30m 04s\n- openstack-ansible-deploy-hosts_lxc_overlayfs-debian-bookworm https://zuul.opendev.org/t/openstack/build/ae6e0eb48ec24de788d49650257cf055 : SUCCESS in 31m 08s\n- openstack-ansible-deploy-hosts_lxc_overlayfs-rockylinux-10 https://zuul.opendev.org/t/openstack/build/a13add5cf1bb41a4b31c2756bff7dd2c : SUCCESS in 29m 44s\n- openstack-ansible-deploy-hosts_lxc_btrfs-debian-trixie https://zuul.opendev.org/t/openstack/build/a21f102751d34d279917e6ade2bee738 : SUCCESS in 31m 52s\n- openstack-ansible-deploy-hosts_lxc_overlayfs-debian-trixie https://zuul.opendev.org/t/openstack/build/455062abadee469aa6ba059b4c406b5b : SUCCESS in 29m 29s","accounts_in_message":[],"_revision_number":4},{"id":"6f1c241ab73c537f7950f307039215a0a10cf616","author":{"_account_id":25023,"name":"Jonathan Rosser","email":"jonathan.rosser@rd.bbc.co.uk","username":"jrosser"},"date":"2026-05-13 08:38:54.000000000","message":"Patch Set 4: Code-Review+2","accounts_in_message":[],"_revision_number":4},{"id":"9e3fea33a3805ddc58b7b2de439495058aca250b","author":{"_account_id":32666,"name":"Damian Dąbrowski","email":"damian@dabrowski.cloud","username":"ddabrowski"},"date":"2026-05-14 11:27:26.000000000","message":"Patch Set 4: Code-Review+2 Workflow+1","accounts_in_message":[],"_revision_number":4},{"id":"1e4eb3f5e7eb8ba17a0b07a5d2895de25a41d0e9","tag":"autogenerated:zuul:gate","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2026-05-14 11:30:07.000000000","message":"Patch Set 4: -Verified\n\nStarting gate jobs.","accounts_in_message":[],"_revision_number":4},{"id":"988da7321ab2d5f45b9ad4acc90e1d1e35e53801","tag":"autogenerated:zuul:gate","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2026-05-14 14:18:59.000000000","message":"Patch Set 4: Verified-2\n\nBuild failed (gate pipeline).  For information on how to proceed, see\nhttps://docs.opendev.org/opendev/infra-manual/latest/developers.html#automated-testing\n\nhttps://zuul.opendev.org/t/openstack/buildset/be9f80be1a0949ac9a1945006ae5cced\n\n- openstack-ansible-linters-ubuntu-noble https://zuul.opendev.org/t/openstack/build/2cb5163bd24940a7b04fd9a3a3d23be9 : SUCCESS in 13m 23s\n- openstack-tox-docs https://zuul.opendev.org/t/openstack/build/65e0ebd92df1497fa6cafa9966f333b7 : SUCCESS in 6m 21s\n- build-openstack-releasenotes https://zuul.opendev.org/t/openstack/build/45bfa660d11442d9a9ffffc966981079 : SUCCESS in 2m 50s\n- openstack-ansible-deploy-infra_lxc-centos-10-stream https://zuul.opendev.org/t/openstack/build/94a4fd10e4f54f2e9e1a303347f24ae4 : SUCCESS in 2h 08m 49s\n- openstack-ansible-deploy-infra_lxc-debian-bookworm https://zuul.opendev.org/t/openstack/build/c9a4afd2e7b34d64b2b2caf3a5431a3c : SUCCESS in 1h 49m 43s\n- openstack-ansible-deploy-infra_lxc-debian-trixie https://zuul.opendev.org/t/openstack/build/b46920f59df244288145bb5c687437f7 : SUCCESS in 2h 00m 58s\n- openstack-ansible-deploy-infra_lxc-rockylinux-10 https://zuul.opendev.org/t/openstack/build/4c0dbdb944724bca89a970e17d0849ee : FAILURE in 1h 30m 06s\n- openstack-ansible-deploy-infra_lxc_tls-rockylinux-10 https://zuul.opendev.org/t/openstack/build/62cdbec8bc5d4c678f541de288dd05af : SUCCESS in 2h 26m 15s\n- openstack-ansible-deploy-infra_lxc-ubuntu-noble https://zuul.opendev.org/t/openstack/build/f1d140fd13cb4c77880a26c4f7355cd5 : SUCCESS in 1h 56m 13s\n- openstack-ansible-deploy-infra_lxc_stepca-ubuntu-noble https://zuul.opendev.org/t/openstack/build/168111b2572b4d8cb0c61530ff9d0f61 : SUCCESS in 1h 56m 29s\n- openstack-ansible-deploy-infra_lxc_tls-ubuntu-noble https://zuul.opendev.org/t/openstack/build/9bcd110721d9418282751f86e06b15df : SUCCESS in 1h 57m 21s\n- openstack-ansible-upgrade-infra_lxc-ubuntu-noble https://zuul.opendev.org/t/openstack/build/8d05e0cc65254c8abd58efc13488d5d6 : SUCCESS in 2h 25m 54s\n- openstack-ansible-deploy-hosts_lxc_zfs-ubuntu-noble https://zuul.opendev.org/t/openstack/build/091cf3565a624e5c9295e3413f401487 : SUCCESS in 31m 42s\n- openstack-ansible-deploy-hosts_lxc_btrfs-ubuntu-noble https://zuul.opendev.org/t/openstack/build/4f2d36b650fb4288862c397dc6b985ca : SUCCESS in 31m 36s\n- openstack-ansible-deploy-hosts_lxc_overlayfs-ubuntu-noble https://zuul.opendev.org/t/openstack/build/7424f9658c9e4b0a8456cc21191984ea : SUCCESS in 29m 55s\n- openstack-ansible-deploy-hosts_lxc_btrfs-debian-bookworm https://zuul.opendev.org/t/openstack/build/a83fd80ed74845a394b945766c93445c : SUCCESS in 28m 58s\n- openstack-ansible-deploy-hosts_lxc_overlayfs-debian-bookworm https://zuul.opendev.org/t/openstack/build/3d788e9b8f014a3ba390dd4932380511 : SUCCESS in 31m 46s\n- openstack-ansible-deploy-hosts_lxc_btrfs-debian-trixie https://zuul.opendev.org/t/openstack/build/63bf3a3c62104193a75de9f9c3337fef : SUCCESS in 31m 39s\n- openstack-ansible-deploy-hosts_lxc_overlayfs-debian-trixie https://zuul.opendev.org/t/openstack/build/9a75e81c218a43d6b43e4fb9f840805b : SUCCESS in 30m 09s\n- openstack-ansible-deploy-hosts_lxc_overlayfs-rockylinux-10 https://zuul.opendev.org/t/openstack/build/0474dd34352e43418ffbdb56c6e89795 : SUCCESS in 28m 41s","accounts_in_message":[],"_revision_number":4},{"id":"bb8377f3a7a150597afeb1ab412e48af5dd93927","author":{"_account_id":28619,"name":"Dmitriy Rabotyagov","email":"noonedeadpunk@gmail.com","username":"noonedeadpunk"},"date":"2026-05-14 14:22:20.000000000","message":"Patch Set 4:\n\n(1 comment)","accounts_in_message":[],"_revision_number":4},{"id":"3a5469000b37a19ce3809134794aca73dd5468a9","tag":"autogenerated:zuul:check","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2026-05-14 17:23:45.000000000","message":"Patch Set 4: Verified-1\n\nBuild failed (check pipeline).  For information on how to proceed, see\nhttps://docs.opendev.org/opendev/infra-manual/latest/developers.html#automated-testing\nand https://docs.openstack.org/project-team-guide/testing.html#how-to-handle-test-failures\n\nhttps://zuul.opendev.org/t/openstack/buildset/c475a625450746a1882424171be675ef\n\n- openstack-ansible-linters-ubuntu-noble https://zuul.opendev.org/t/openstack/build/a508a2f27aa54ebea98e76a4ab4e81f4 : SUCCESS in 15m 59s\n- openstack-tox-docs https://zuul.opendev.org/t/openstack/build/28e14028da2647caad8f1128f9eaad54 : SUCCESS in 6m 25s\n- build-openstack-releasenotes https://zuul.opendev.org/t/openstack/build/4bb8f1a2f54347ea8c49afcf05b6a8e3 : SUCCESS in 3m 37s\n- openstack-ansible-deploy-infra_lxc-centos-10-stream https://zuul.opendev.org/t/openstack/build/8f18dfb668304201b49c529bceb942d4 : SUCCESS in 1h 12m 16s\n- openstack-ansible-deploy-infra_lxc-debian-bookworm https://zuul.opendev.org/t/openstack/build/15660cfebad54e9e9c0deb2bd8ade3cb : SUCCESS in 1h 56m 21s\n- openstack-ansible-deploy-infra_lxc-debian-trixie https://zuul.opendev.org/t/openstack/build/3b401f073f2b4fcf9ef496ebe36699cc : SUCCESS in 2h 00m 21s\n- openstack-ansible-deploy-infra_lxc-rockylinux-10 https://zuul.opendev.org/t/openstack/build/ccdf8c248a5348b0a721aeed17e5d14d : SUCCESS in 1h 57m 33s\n- openstack-ansible-deploy-infra_lxc_tls-rockylinux-10 https://zuul.opendev.org/t/openstack/build/c2c26a2168da4c77b7a2f0a38b66097f : SUCCESS in 1h 12m 31s\n- openstack-ansible-deploy-infra_lxc-ubuntu-noble https://zuul.opendev.org/t/openstack/build/b53b35f153104e839f7fb37686b59d04 : FAILURE in 1h 12m 34s\n- openstack-ansible-deploy-infra_lxc_stepca-ubuntu-noble https://zuul.opendev.org/t/openstack/build/9ef87b8530a2445f9348fdf12ec5ea95 : SUCCESS in 1h 46m 34s\n- openstack-ansible-deploy-infra_lxc_tls-ubuntu-noble https://zuul.opendev.org/t/openstack/build/a1dc7a3857a84a52a26f28cff9fbdd2b : FAILURE in 1h 15m 50s\n- openstack-ansible-upgrade-infra_lxc-ubuntu-noble https://zuul.opendev.org/t/openstack/build/f6f6b846c6da4bd586407c381d6dcf1e : SUCCESS in 2h 36m 03s\n- openstack-ansible-deploy-hosts_lxc_zfs-ubuntu-noble https://zuul.opendev.org/t/openstack/build/2ac61bfca6994dea9b4e2f0081fa639b : SUCCESS in 30m 55s\n- openstack-ansible-deploy-hosts_lxc_btrfs-ubuntu-noble https://zuul.opendev.org/t/openstack/build/f16065118f2346d5b2e096fd8c7a5499 : SUCCESS in 32m 21s\n- openstack-ansible-deploy-hosts_lxc_overlayfs-ubuntu-noble https://zuul.opendev.org/t/openstack/build/ab127bdd67c94b7996e541313cd9f773 : SUCCESS in 32m 40s\n- openstack-ansible-deploy-hosts_lxc_btrfs-debian-bookworm https://zuul.opendev.org/t/openstack/build/33460f9d8dda4be2ab32d6b7d10c8ce1 : SUCCESS in 35m 01s\n- openstack-ansible-deploy-hosts_lxc_overlayfs-debian-bookworm https://zuul.opendev.org/t/openstack/build/4b93c4b761ac4ca5b8bb3845b6527194 : SUCCESS in 31m 10s\n- openstack-ansible-deploy-hosts_lxc_overlayfs-rockylinux-10 https://zuul.opendev.org/t/openstack/build/be130fee3d1d456aac84df4222567b41 : SUCCESS in 37m 32s\n- openstack-ansible-deploy-hosts_lxc_btrfs-debian-trixie https://zuul.opendev.org/t/openstack/build/cccd393dccf0424993ecc4c4aa89d631 : SUCCESS in 30m 14s\n- openstack-ansible-deploy-hosts_lxc_overlayfs-debian-trixie https://zuul.opendev.org/t/openstack/build/d25b24f5d90b40b99b1dc610b27095d9 : SUCCESS in 30m 12s","accounts_in_message":[],"_revision_number":4},{"id":"867daade1793afdba8a3dafaa1de13fe91321243","author":{"_account_id":37598,"name":"Ivan Anfimov","display_name":"Ivan Anfimov","email":"lazekteam@gmail.com","username":"anfimovir"},"date":"2026-05-16 18:14:44.000000000","message":"Patch Set 4:\n\n(1 comment)","accounts_in_message":[],"_revision_number":4},{"id":"8d4ad1463bb971db778446714b072439ecff2584","tag":"autogenerated:zuul:check","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2026-05-16 20:19:14.000000000","message":"Patch Set 4:\n\nBuild failed (check pipeline).  For information on how to proceed, see\nhttps://docs.opendev.org/opendev/infra-manual/latest/developers.html#automated-testing\nand https://docs.openstack.org/project-team-guide/testing.html#how-to-handle-test-failures\n\nhttps://zuul.opendev.org/t/openstack/buildset/e73ac3ec38cb43d69802cf372ec5bbf1\n\n- openstack-ansible-linters-ubuntu-noble https://zuul.opendev.org/t/openstack/build/066dba07d431475e903353b57e33e92e : SUCCESS in 10m 20s\n- openstack-tox-docs https://zuul.opendev.org/t/openstack/build/7f4d549f653a45c2adf7ee690d0b778b : SUCCESS in 4m 51s\n- build-openstack-releasenotes https://zuul.opendev.org/t/openstack/build/ab5e057a4d8041b99a714d12a0ba6634 : SUCCESS in 4m 19s\n- openstack-ansible-deploy-infra_lxc-centos-10-stream https://zuul.opendev.org/t/openstack/build/82e62bfb04a2416998f5aefeb6e4e085 : SUCCESS in 1h 54m 46s\n- openstack-ansible-deploy-infra_lxc-debian-bookworm https://zuul.opendev.org/t/openstack/build/59dd2ff6a345409982785eda487fca47 : SUCCESS in 57m 31s\n- openstack-ansible-deploy-infra_lxc-debian-trixie https://zuul.opendev.org/t/openstack/build/4daa9c376376487cad2b821b91fae0ec : SUCCESS in 1h 47m 55s\n- openstack-ansible-deploy-infra_lxc-rockylinux-10 https://zuul.opendev.org/t/openstack/build/3f44901a1c004469b0961d0d12c60ed2 : FAILURE in 1h 38m 39s\n- openstack-ansible-deploy-infra_lxc_tls-rockylinux-10 https://zuul.opendev.org/t/openstack/build/e9e35ac6b0b34a54ae4a133d2f28423e : SUCCESS in 1h 36m 17s\n- openstack-ansible-deploy-infra_lxc-ubuntu-noble https://zuul.opendev.org/t/openstack/build/84395ef4b67243828854474facc11926 : SUCCESS in 1h 08m 07s\n- openstack-ansible-deploy-infra_lxc_stepca-ubuntu-noble https://zuul.opendev.org/t/openstack/build/3bbb4c6b1e76418a8a51d2291020f069 : SUCCESS in 1h 52m 36s\n- openstack-ansible-deploy-infra_lxc_tls-ubuntu-noble https://zuul.opendev.org/t/openstack/build/64b47ef0744c47a9bdaf27bd5aae234a : SUCCESS in 1h 48m 39s\n- openstack-ansible-upgrade-infra_lxc-ubuntu-noble https://zuul.opendev.org/t/openstack/build/93bbaeb5ff6c42aab006f29ea7c51c8b : SUCCESS in 1h 06m 41s\n- openstack-ansible-deploy-hosts_lxc_zfs-ubuntu-noble https://zuul.opendev.org/t/openstack/build/b2098a89e7994680a0441977beaafa68 : SUCCESS in 23m 04s\n- openstack-ansible-deploy-hosts_lxc_btrfs-ubuntu-noble https://zuul.opendev.org/t/openstack/build/0f196972ac3c40c2bac222914bc76cae : SUCCESS in 19m 56s\n- openstack-ansible-deploy-hosts_lxc_overlayfs-ubuntu-noble https://zuul.opendev.org/t/openstack/build/e52a76aa02f64560806281a7eb2ba6cf : SUCCESS in 32m 14s\n- openstack-ansible-deploy-hosts_lxc_btrfs-debian-bookworm https://zuul.opendev.org/t/openstack/build/674429dcfa6b48119cbd64536b6caa73 : SUCCESS in 29m 04s\n- openstack-ansible-deploy-hosts_lxc_overlayfs-debian-bookworm https://zuul.opendev.org/t/openstack/build/e5803f4f28a849bdbc91b63d2ce1cc52 : SUCCESS in 28m 06s\n- openstack-ansible-deploy-hosts_lxc_overlayfs-rockylinux-10 https://zuul.opendev.org/t/openstack/build/442109ce2a344a9aa53d6b90387c3aa4 : SUCCESS in 19m 57s\n- openstack-ansible-deploy-hosts_lxc_btrfs-debian-trixie https://zuul.opendev.org/t/openstack/build/aa94a911668b45e0b49178a384370808 : SUCCESS in 28m 04s\n- openstack-ansible-deploy-hosts_lxc_overlayfs-debian-trixie https://zuul.opendev.org/t/openstack/build/ccba28bcda264d5795c1b03e30489676 : SUCCESS in 19m 51s","accounts_in_message":[],"_revision_number":4},{"id":"919afe743587f64a671a8fb49310c352f6dc0bc3","author":{"_account_id":28619,"name":"Dmitriy Rabotyagov","email":"noonedeadpunk@gmail.com","username":"noonedeadpunk"},"date":"2026-05-19 15:10:39.000000000","message":"Patch Set 4:\n\n(1 comment)","accounts_in_message":[],"_revision_number":4},{"id":"3e55a1442db8f37319fd5aeb66080aa1aa2e5850","tag":"autogenerated:zuul:check","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2026-05-19 17:48:05.000000000","message":"Patch Set 4: Verified+1\n\nBuild succeeded (check pipeline).\nhttps://zuul.opendev.org/t/openstack/buildset/2619e5a9c81047469afb519acabdbc4c\n\n- openstack-ansible-linters-ubuntu-noble https://zuul.opendev.org/t/openstack/build/24ecfb27304342179b8a419ac8ace677 : SUCCESS in 14m 02s\n- openstack-tox-docs https://zuul.opendev.org/t/openstack/build/0a1d6190e7fc487694b93ef6e8ad76ad : SUCCESS in 3m 54s\n- build-openstack-releasenotes https://zuul.opendev.org/t/openstack/build/58ac0443f22345b894f324ec9ab767c1 : SUCCESS in 3m 39s\n- openstack-ansible-deploy-infra_lxc-centos-10-stream https://zuul.opendev.org/t/openstack/build/94a201ab1c4046a0a3ed905912d95ccb : SUCCESS in 1h 14m 24s\n- openstack-ansible-deploy-infra_lxc-debian-bookworm https://zuul.opendev.org/t/openstack/build/2714e3b0260644e2afff81b1bf6e2192 : SUCCESS in 1h 42m 45s\n- openstack-ansible-deploy-infra_lxc-debian-trixie https://zuul.opendev.org/t/openstack/build/98ac14410b104f4e90fce5290a6c217f : SUCCESS in 1h 12m 38s\n- openstack-ansible-deploy-infra_lxc-rockylinux-10 https://zuul.opendev.org/t/openstack/build/f54f9c01b0854c2bb0e639d628113324 : SUCCESS in 2h 12m 38s\n- openstack-ansible-deploy-infra_lxc_tls-rockylinux-10 https://zuul.opendev.org/t/openstack/build/465f464f445f496aa857f2e6b9c272fb : SUCCESS in 2h 12m 48s\n- openstack-ansible-deploy-infra_lxc-ubuntu-noble https://zuul.opendev.org/t/openstack/build/751a473534fe42a9af5f933d9438bb34 : SUCCESS in 1h 50m 50s\n- openstack-ansible-deploy-infra_lxc_stepca-ubuntu-noble https://zuul.opendev.org/t/openstack/build/0e9675cf9b6941c8b497e2f2cb662278 : SUCCESS in 1h 52m 10s\n- openstack-ansible-deploy-infra_lxc_tls-ubuntu-noble https://zuul.opendev.org/t/openstack/build/dc9df05ec316449faf9bd201272c75fb : SUCCESS in 2h 02m 31s\n- openstack-ansible-upgrade-infra_lxc-ubuntu-noble https://zuul.opendev.org/t/openstack/build/ccc348e2264f44d5a93d7e7a60f494b1 : SUCCESS in 2h 25m 39s\n- openstack-ansible-deploy-hosts_lxc_zfs-ubuntu-noble https://zuul.opendev.org/t/openstack/build/253289282a8c48c09267a44c9b0b4c89 : SUCCESS in 40m 23s\n- openstack-ansible-deploy-hosts_lxc_btrfs-ubuntu-noble https://zuul.opendev.org/t/openstack/build/5bd360f1fc0b444bb16912d43fcb94e5 : SUCCESS in 30m 51s\n- openstack-ansible-deploy-hosts_lxc_overlayfs-ubuntu-noble https://zuul.opendev.org/t/openstack/build/1aa2638583e64713af80e70abf4d72ba : SUCCESS in 28m 21s\n- openstack-ansible-deploy-hosts_lxc_btrfs-debian-bookworm https://zuul.opendev.org/t/openstack/build/91fdd9955e3a43f094d6178f0c99963c : SUCCESS in 29m 52s\n- openstack-ansible-deploy-hosts_lxc_overlayfs-debian-bookworm https://zuul.opendev.org/t/openstack/build/0c409df649c846988248e89e06a0cfca : SUCCESS in 30m 33s\n- openstack-ansible-deploy-hosts_lxc_overlayfs-rockylinux-10 https://zuul.opendev.org/t/openstack/build/39bfa4329b99469c8e82ecaac2f33c58 : SUCCESS in 30m 10s\n- openstack-ansible-deploy-hosts_lxc_btrfs-debian-trixie https://zuul.opendev.org/t/openstack/build/564d748c87f541fcac330c510d0d8266 : SUCCESS in 27m 32s\n- openstack-ansible-deploy-hosts_lxc_overlayfs-debian-trixie https://zuul.opendev.org/t/openstack/build/00056f384bb349eeb0860448ceb3089b : SUCCESS in 32m 00s","accounts_in_message":[],"_revision_number":4},{"id":"5314ab34d349226c85ad109207d97b54613664aa","tag":"autogenerated:zuul:gate","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2026-05-19 17:50:51.000000000","message":"Patch Set 4: -Verified\n\nStarting gate jobs.","accounts_in_message":[],"_revision_number":4},{"id":"cff68cbcd115b2255119b182810b3b15a8c381ce","tag":"autogenerated:zuul:gate","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2026-05-19 19:12:45.000000000","message":"Patch Set 4: Verified-2\n\nBuild failed (gate pipeline).  For information on how to proceed, see\nhttps://docs.opendev.org/opendev/infra-manual/latest/developers.html#automated-testing\n\nhttps://zuul.opendev.org/t/openstack/buildset/a118a114d8e64961beba5a538d4bdafc\n\n- openstack-ansible-linters-ubuntu-noble https://zuul.opendev.org/t/openstack/build/9f17ae78407c4fb6aa71c853308dec0a : SUCCESS in 13m 13s\n- openstack-tox-docs https://zuul.opendev.org/t/openstack/build/0a5a2afdfbe14ac08a549f28b76fedfc : SUCCESS in 7m 30s\n- build-openstack-releasenotes https://zuul.opendev.org/t/openstack/build/a89eee343d754e66b2111da57f8c82f8 : SUCCESS in 4m 01s\n- openstack-ansible-deploy-infra_lxc-centos-10-stream https://zuul.opendev.org/t/openstack/build/806263bfcc274926934aa8f6378c40d5 : FAILURE in 28m 14s\n- openstack-ansible-deploy-infra_lxc-debian-bookworm https://zuul.opendev.org/t/openstack/build/6654884291fa47209d6838b7dbb2e69e : FAILURE in 39m 37s\n- openstack-ansible-deploy-infra_lxc-debian-trixie https://zuul.opendev.org/t/openstack/build/7eade0559fec4dd3b78de90051c9e884 : FAILURE in 31m 32s\n- openstack-ansible-deploy-infra_lxc-rockylinux-10 https://zuul.opendev.org/t/openstack/build/1f6083247b1746fcb1ae2828f85e28d2 : FAILURE in 42m 10s\n- openstack-ansible-deploy-infra_lxc_tls-rockylinux-10 https://zuul.opendev.org/t/openstack/build/a84d8dd8650140b6837c05a75531d2fe : FAILURE in 31m 15s\n- openstack-ansible-deploy-infra_lxc-ubuntu-noble https://zuul.opendev.org/t/openstack/build/edde66720096491ab3352421db895dbd : FAILURE in 34m 53s\n- openstack-ansible-deploy-infra_lxc_stepca-ubuntu-noble https://zuul.opendev.org/t/openstack/build/084d4070e7684bb1aac2f738b870266f : FAILURE in 36m 35s\n- openstack-ansible-deploy-infra_lxc_tls-ubuntu-noble https://zuul.opendev.org/t/openstack/build/1c692e6f379b40c4acbff6f2958ef64c : FAILURE in 35m 56s\n- openstack-ansible-upgrade-infra_lxc-ubuntu-noble https://zuul.opendev.org/t/openstack/build/cbaaa073b9ff4e72baa1b22e344e0205 : FAILURE in 1h 14m 01s\n- openstack-ansible-deploy-hosts_lxc_zfs-ubuntu-noble https://zuul.opendev.org/t/openstack/build/3b4058e1dca5482d8aa4eec60c74f882 : SUCCESS in 29m 38s\n- openstack-ansible-deploy-hosts_lxc_btrfs-ubuntu-noble https://zuul.opendev.org/t/openstack/build/f52ec457662e4977acfe7e1a45d9c774 : SUCCESS in 27m 00s\n- openstack-ansible-deploy-hosts_lxc_overlayfs-ubuntu-noble https://zuul.opendev.org/t/openstack/build/aa6d2d06b5e84c0ea25a94256780f420 : SUCCESS in 20m 23s\n- openstack-ansible-deploy-hosts_lxc_btrfs-debian-bookworm https://zuul.opendev.org/t/openstack/build/2b9ad096c15041b589e0380f8984d4e2 : SUCCESS in 29m 25s\n- openstack-ansible-deploy-hosts_lxc_overlayfs-debian-bookworm https://zuul.opendev.org/t/openstack/build/f6f57ed1157c4c71aea7c49f1e8b8340 : SUCCESS in 30m 46s\n- openstack-ansible-deploy-hosts_lxc_btrfs-debian-trixie https://zuul.opendev.org/t/openstack/build/25d36989670c43ecb8b7d139f5d83d8e : SUCCESS in 27m 04s\n- openstack-ansible-deploy-hosts_lxc_overlayfs-debian-trixie https://zuul.opendev.org/t/openstack/build/d2e1d9a2bee34c5d995f09bb55b54fbc : SUCCESS in 33m 26s\n- openstack-ansible-deploy-hosts_lxc_overlayfs-rockylinux-10 https://zuul.opendev.org/t/openstack/build/01b3d86a01694a6da6b5269ba49c27c1 : SUCCESS in 27m 50s","accounts_in_message":[],"_revision_number":4},{"id":"3d8c630e1e83bde68ae323ff110814fc930a3c1b","author":{"_account_id":28619,"name":"Dmitriy Rabotyagov","email":"noonedeadpunk@gmail.com","username":"noonedeadpunk"},"date":"2026-05-20 14:23:50.000000000","message":"Patch Set 4:\n\n(1 comment)","accounts_in_message":[],"_revision_number":4},{"id":"370941d13e89621c3b577de84fad9c4de5dcad45","tag":"autogenerated:zuul:check","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2026-05-20 16:33:25.000000000","message":"Patch Set 4: Verified-1\n\nBuild failed (check pipeline).  For information on how to proceed, see\nhttps://docs.opendev.org/opendev/infra-manual/latest/developers.html#automated-testing\nand https://docs.openstack.org/project-team-guide/testing.html#how-to-handle-test-failures\n\nhttps://zuul.opendev.org/t/openstack/buildset/5f39703fffda4a418471f563d824ddfe\n\n- openstack-ansible-linters-ubuntu-noble https://zuul.opendev.org/t/openstack/build/e07e4eb70ae147b8804b1ad527b3620e : SUCCESS in 14m 04s\n- openstack-tox-docs https://zuul.opendev.org/t/openstack/build/0326f85660cf4c99a8ff73e1139b87a5 : SUCCESS in 5m 08s\n- build-openstack-releasenotes https://zuul.opendev.org/t/openstack/build/df5c5de4e129480682d6aea0285daf87 : SUCCESS in 2m 06s\n- openstack-ansible-deploy-infra_lxc-centos-10-stream https://zuul.opendev.org/t/openstack/build/2db98cd4fd0e4463be9486bb6915d677 : FAILURE in 52m 52s\n- openstack-ansible-deploy-infra_lxc-debian-bookworm https://zuul.opendev.org/t/openstack/build/45c7753108f94d589cfef5e51ce0ec9f : SUCCESS in 1h 09m 43s\n- openstack-ansible-deploy-infra_lxc-debian-trixie https://zuul.opendev.org/t/openstack/build/9690f90a0d2a482f9180d403550cdc34 : SUCCESS in 1h 10m 11s\n- openstack-ansible-deploy-infra_lxc-rockylinux-10 https://zuul.opendev.org/t/openstack/build/42e55aaa5c4541bc9c24d564fc651dc6 : SUCCESS in 1h 19m 06s\n- openstack-ansible-deploy-infra_lxc_tls-rockylinux-10 https://zuul.opendev.org/t/openstack/build/7ab02d9d02bd4632a006493c331bcda8 : SUCCESS in 1h 21m 44s\n- openstack-ansible-deploy-infra_lxc-ubuntu-noble https://zuul.opendev.org/t/openstack/build/6a9fad28fa20415d86a164329717fe67 : FAILURE in 1h 26m 12s\n- openstack-ansible-deploy-infra_lxc_stepca-ubuntu-noble https://zuul.opendev.org/t/openstack/build/beee923f3d3843ffa8b6cb937568eed1 : SUCCESS in 1h 59m 06s\n- openstack-ansible-deploy-infra_lxc_tls-ubuntu-noble https://zuul.opendev.org/t/openstack/build/404900d0736d403cb9fb4c307c5722d5 : SUCCESS in 1h 48m 25s\n- openstack-ansible-upgrade-infra_lxc-ubuntu-noble https://zuul.opendev.org/t/openstack/build/f772fa74c3834ef2894934005f3da359 : SUCCESS in 2h 04m 41s\n- openstack-ansible-deploy-hosts_lxc_zfs-ubuntu-noble https://zuul.opendev.org/t/openstack/build/f73cd5c9d5c343aebf56d3ab51210748 : SUCCESS in 29m 22s\n- openstack-ansible-deploy-hosts_lxc_btrfs-ubuntu-noble https://zuul.opendev.org/t/openstack/build/0488ab23b66d4bdcbe51bf8305c3f25f : SUCCESS in 33m 11s\n- openstack-ansible-deploy-hosts_lxc_overlayfs-ubuntu-noble https://zuul.opendev.org/t/openstack/build/428068c2a49b4fe7a9abb65ec2cbde62 : SUCCESS in 28m 57s\n- openstack-ansible-deploy-hosts_lxc_btrfs-debian-bookworm https://zuul.opendev.org/t/openstack/build/6843a7a7366145cb871cacb8a386da7b : SUCCESS in 34m 53s\n- openstack-ansible-deploy-hosts_lxc_overlayfs-debian-bookworm https://zuul.opendev.org/t/openstack/build/ecd2e2e7c05c43f3af8e55319abb9f4e : SUCCESS in 31m 23s\n- openstack-ansible-deploy-hosts_lxc_overlayfs-rockylinux-10 https://zuul.opendev.org/t/openstack/build/947e4b3d7abc4ac88bf014d8b177834f : SUCCESS in 36m 27s\n- openstack-ansible-deploy-hosts_lxc_btrfs-debian-trixie https://zuul.opendev.org/t/openstack/build/15b1491871f744c885f17ff8d0da8d09 : SUCCESS in 22m 16s\n- openstack-ansible-deploy-hosts_lxc_overlayfs-debian-trixie https://zuul.opendev.org/t/openstack/build/fc6c6dc44c7f48caaf3785539270b66c : SUCCESS in 22m 32s","accounts_in_message":[],"_revision_number":4},{"id":"a264da3a6399e1129d27fc7e2b6b75de7daa6481","author":{"_account_id":28619,"name":"Dmitriy Rabotyagov","email":"noonedeadpunk@gmail.com","username":"noonedeadpunk"},"date":"2026-05-21 16:13:40.000000000","message":"Patch Set 4:\n\n(1 comment)","accounts_in_message":[],"_revision_number":4},{"id":"26fc19668c4986a79a3f5229bf3b3cb7f6d6d2be","tag":"autogenerated:zuul:check","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2026-05-21 18:43:42.000000000","message":"Patch Set 4:\n\nBuild failed (check pipeline).  For information on how to proceed, see\nhttps://docs.opendev.org/opendev/infra-manual/latest/developers.html#automated-testing\nand https://docs.openstack.org/project-team-guide/testing.html#how-to-handle-test-failures\n\nhttps://zuul.opendev.org/t/openstack/buildset/4cc1bcd52d10452e87f4ea296fe9ec42\n\n- openstack-ansible-linters-ubuntu-noble https://zuul.opendev.org/t/openstack/build/2a9f1dc766c74d8cbf97cad1f6a27c6b : SUCCESS in 11m 26s\n- openstack-tox-docs https://zuul.opendev.org/t/openstack/build/54d8c130a058411f995160d477945bb7 : SUCCESS in 6m 13s\n- build-openstack-releasenotes https://zuul.opendev.org/t/openstack/build/9bb07002123f4109a3438f736016b7c2 : SUCCESS in 5m 23s\n- openstack-ansible-deploy-infra_lxc-centos-10-stream https://zuul.opendev.org/t/openstack/build/8067529c78ff4ab58c5e93c3f48be673 : SUCCESS in 1h 20m 31s\n- openstack-ansible-deploy-infra_lxc-debian-bookworm https://zuul.opendev.org/t/openstack/build/bf799c7af55f46639a468a99390c7355 : SUCCESS in 1h 08m 41s\n- openstack-ansible-deploy-infra_lxc-debian-trixie https://zuul.opendev.org/t/openstack/build/31066d5852bb48daa8d4d4538ba8bd27 : SUCCESS in 1h 59m 08s\n- openstack-ansible-deploy-infra_lxc-rockylinux-10 https://zuul.opendev.org/t/openstack/build/0a265db3b32a4faaaa8b1bc089c184ca : SUCCESS in 2h 10m 48s\n- openstack-ansible-deploy-infra_lxc_tls-rockylinux-10 https://zuul.opendev.org/t/openstack/build/5fc6e56b113b44479e50df000e0fd29a : FAILURE in 1h 14m 59s\n- openstack-ansible-deploy-infra_lxc-ubuntu-noble https://zuul.opendev.org/t/openstack/build/07af6695498a4446aaffec98c8093385 : SUCCESS in 2h 13m 44s\n- openstack-ansible-deploy-infra_lxc_stepca-ubuntu-noble https://zuul.opendev.org/t/openstack/build/6c45b70b7f3f4ca8b1c62a4ba2f9f40d : SUCCESS in 2h 14m 43s\n- openstack-ansible-deploy-infra_lxc_tls-ubuntu-noble https://zuul.opendev.org/t/openstack/build/b07d41a02e9443d5862e884a37d9b249 : SUCCESS in 1h 04m 53s\n- openstack-ansible-upgrade-infra_lxc-ubuntu-noble https://zuul.opendev.org/t/openstack/build/f7803b8eceb04222a8bfe1303c23aad8 : SUCCESS in 1h 36m 38s\n- openstack-ansible-deploy-hosts_lxc_zfs-ubuntu-noble https://zuul.opendev.org/t/openstack/build/69397edc1662497b862461a6319d0a7c : SUCCESS in 28m 41s\n- openstack-ansible-deploy-hosts_lxc_btrfs-ubuntu-noble https://zuul.opendev.org/t/openstack/build/8109f7d21a8540dd819a4a5a67d82d37 : SUCCESS in 30m 31s\n- openstack-ansible-deploy-hosts_lxc_overlayfs-ubuntu-noble https://zuul.opendev.org/t/openstack/build/cdbd9255e75f477eb9af11e10942fdbf : SUCCESS in 27m 07s\n- openstack-ansible-deploy-hosts_lxc_btrfs-debian-bookworm https://zuul.opendev.org/t/openstack/build/2bc76b14515d47c7858dd3743eeb72e8 : SUCCESS in 31m 12s\n- openstack-ansible-deploy-hosts_lxc_overlayfs-debian-bookworm https://zuul.opendev.org/t/openstack/build/116c6437aca5478484ab3a9166decb44 : SUCCESS in 25m 30s\n- openstack-ansible-deploy-hosts_lxc_overlayfs-rockylinux-10 https://zuul.opendev.org/t/openstack/build/6bdfa1f32fcd48cba9e1ce678102c7e3 : SUCCESS in 28m 52s\n- openstack-ansible-deploy-hosts_lxc_btrfs-debian-trixie https://zuul.opendev.org/t/openstack/build/5d25da0f96f54e4cbec5f3161ae8e031 : SUCCESS in 27m 23s\n- openstack-ansible-deploy-hosts_lxc_overlayfs-debian-trixie https://zuul.opendev.org/t/openstack/build/4c9b188f98564935a0f43b60ab1af46d : SUCCESS in 23m 17s","accounts_in_message":[],"_revision_number":4},{"id":"c40b5769cfabfa6095428f31a2572598794dccdc","author":{"_account_id":37598,"name":"Ivan Anfimov","display_name":"Ivan Anfimov","email":"lazekteam@gmail.com","username":"anfimovir"},"date":"2026-05-24 21:13:49.000000000","message":"Patch Set 4:\n\n(1 comment)","accounts_in_message":[],"_revision_number":4},{"id":"10e5142bbd51505b69d853879c6bf95c2e1a8b81","tag":"autogenerated:zuul:check","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2026-05-24 23:33:48.000000000","message":"Patch Set 4: Verified+1\n\nBuild succeeded (check pipeline).\nhttps://zuul.opendev.org/t/openstack/buildset/4e1ffec8bcd04e2cbb2443c8394b96a2\n\n- openstack-ansible-linters-ubuntu-noble https://zuul.opendev.org/t/openstack/build/0207008f7efc4283934bb01588e7c53a : SUCCESS in 16m 00s\n- openstack-tox-docs https://zuul.opendev.org/t/openstack/build/6d56dd67258a4d00831ddacf9c59f2e7 : SUCCESS in 7m 07s\n- build-openstack-releasenotes https://zuul.opendev.org/t/openstack/build/418c18dead8b4f55937ca6592a916768 : SUCCESS in 4m 06s\n- openstack-ansible-deploy-infra_lxc-centos-10-stream https://zuul.opendev.org/t/openstack/build/6284ec90a8ec4cc281041bc4b97aaef4 : SUCCESS in 1h 25m 30s\n- openstack-ansible-deploy-infra_lxc-debian-bookworm https://zuul.opendev.org/t/openstack/build/15c76076bd8e43d589364e4a323003b1 : SUCCESS in 1h 47m 40s\n- openstack-ansible-deploy-infra_lxc-debian-trixie https://zuul.opendev.org/t/openstack/build/52cabf8ea6c74fe29e0d19c62d86eb32 : SUCCESS in 1h 55m 54s\n- openstack-ansible-deploy-infra_lxc-rockylinux-10 https://zuul.opendev.org/t/openstack/build/38c07748cb76488c88bf180dd4417bd0 : SUCCESS in 1h 20m 41s\n- openstack-ansible-deploy-infra_lxc_tls-rockylinux-10 https://zuul.opendev.org/t/openstack/build/9ea1c25b76fb40ddb33753fdf5dfb563 : SUCCESS in 1h 47m 31s\n- openstack-ansible-deploy-infra_lxc-ubuntu-noble https://zuul.opendev.org/t/openstack/build/bf677a4b1ab84f209d2e16fa8265103f : SUCCESS in 2h 01m 59s\n- openstack-ansible-deploy-infra_lxc_stepca-ubuntu-noble https://zuul.opendev.org/t/openstack/build/ebb1ce30afaa4925ad37b16aa38fbe1a : SUCCESS in 1h 52m 31s\n- openstack-ansible-deploy-infra_lxc_tls-ubuntu-noble https://zuul.opendev.org/t/openstack/build/dc36f100f58946a0be028956c9eed5e4 : SUCCESS in 1h 50m 38s\n- openstack-ansible-upgrade-infra_lxc-ubuntu-noble https://zuul.opendev.org/t/openstack/build/eb33f0ef02a24241ba405f3d8df33900 : SUCCESS in 2h 12m 47s\n- openstack-ansible-deploy-hosts_lxc_zfs-ubuntu-noble https://zuul.opendev.org/t/openstack/build/43a4e57bdb9f42cb80ef6878778b8393 : SUCCESS in 33m 08s\n- openstack-ansible-deploy-hosts_lxc_btrfs-ubuntu-noble https://zuul.opendev.org/t/openstack/build/a44f3ef17de44f02b7eb75997b965998 : SUCCESS in 33m 23s\n- openstack-ansible-deploy-hosts_lxc_overlayfs-ubuntu-noble https://zuul.opendev.org/t/openstack/build/5e69e32046ca4bf991076c94d5f146d9 : SUCCESS in 35m 40s\n- openstack-ansible-deploy-hosts_lxc_btrfs-debian-bookworm https://zuul.opendev.org/t/openstack/build/e5dfd5211f8a45638bd56996da68c45f : SUCCESS in 29m 57s\n- openstack-ansible-deploy-hosts_lxc_overlayfs-debian-bookworm https://zuul.opendev.org/t/openstack/build/68254582abec487898f97905fc53fc17 : SUCCESS in 30m 03s\n- openstack-ansible-deploy-hosts_lxc_overlayfs-rockylinux-10 https://zuul.opendev.org/t/openstack/build/60cd951fa4a8475ba53063544b7261bf : SUCCESS in 29m 14s\n- openstack-ansible-deploy-hosts_lxc_btrfs-debian-trixie https://zuul.opendev.org/t/openstack/build/32772ab2e0f1426d8dbf2912b3279a7a : SUCCESS in 31m 09s\n- openstack-ansible-deploy-hosts_lxc_overlayfs-debian-trixie https://zuul.opendev.org/t/openstack/build/5915526146f04cf6b922a71d39522662 : SUCCESS in 31m 21s","accounts_in_message":[],"_revision_number":4},{"id":"ec3e18eb8a924ed2ca24e133a7c0b77d76292e5e","tag":"autogenerated:zuul:gate","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2026-05-24 23:36:24.000000000","message":"Patch Set 4: -Verified\n\nStarting gate jobs.","accounts_in_message":[],"_revision_number":4},{"id":"1ffbf4b5a7290fe06ba3cb2b00c370440d18b7f6","tag":"autogenerated:zuul:gate","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2026-05-25 02:07:08.000000000","message":"Patch Set 4: Verified-2\n\nBuild failed (gate pipeline).  For information on how to proceed, see\nhttps://docs.opendev.org/opendev/infra-manual/latest/developers.html#automated-testing\n\nhttps://zuul.opendev.org/t/openstack/buildset/abb285f4fd3f42fea49ccf4a80194a91\n\n- openstack-ansible-linters-ubuntu-noble https://zuul.opendev.org/t/openstack/build/f093c6a1e8884e0ab471b8e97c8c84cf : SUCCESS in 7m 40s\n- openstack-tox-docs https://zuul.opendev.org/t/openstack/build/a6f10fdb3d354f97b1fbe8dca42fb116 : SUCCESS in 9m 18s\n- build-openstack-releasenotes https://zuul.opendev.org/t/openstack/build/d1bac8fc6686425ba10a20cf523156b7 : SUCCESS in 2m 01s\n- openstack-ansible-deploy-infra_lxc-centos-10-stream https://zuul.opendev.org/t/openstack/build/fca369d4fffb4a779f2eabbc41261756 : FAILURE in 1h 12m 17s\n- openstack-ansible-deploy-infra_lxc-debian-bookworm https://zuul.opendev.org/t/openstack/build/c712c4f8e3904da08908b035b2072417 : SUCCESS in 58m 24s\n- openstack-ansible-deploy-infra_lxc-debian-trixie https://zuul.opendev.org/t/openstack/build/259b2cf99cec4bb2804bcbb5475b0d33 : SUCCESS in 1h 01m 47s\n- openstack-ansible-deploy-infra_lxc-rockylinux-10 https://zuul.opendev.org/t/openstack/build/935cbf776ea2472d85d43c049f76ef0d : SUCCESS in 2h 05m 27s\n- openstack-ansible-deploy-infra_lxc_tls-rockylinux-10 https://zuul.opendev.org/t/openstack/build/c2c7a646e13a4403ba0c54d5f9bcb4f2 : SUCCESS in 1h 34m 20s\n- openstack-ansible-deploy-infra_lxc-ubuntu-noble https://zuul.opendev.org/t/openstack/build/266dbc62874f4ea586c936e1cc5f3c29 : SUCCESS in 1h 43m 26s\n- openstack-ansible-deploy-infra_lxc_stepca-ubuntu-noble https://zuul.opendev.org/t/openstack/build/336af51b5aa8465da328ecbeb96a9104 : SUCCESS in 1h 50m 37s\n- openstack-ansible-deploy-infra_lxc_tls-ubuntu-noble https://zuul.opendev.org/t/openstack/build/701915509e7d4272905903c28315c907 : SUCCESS in 1h 51m 45s\n- openstack-ansible-upgrade-infra_lxc-ubuntu-noble https://zuul.opendev.org/t/openstack/build/523dae4cef954f7ea1229cd67d2528e6 : SUCCESS in 2h 20m 08s\n- openstack-ansible-deploy-hosts_lxc_zfs-ubuntu-noble https://zuul.opendev.org/t/openstack/build/4917401f03da4eb185e51830c78a121a : SUCCESS in 24m 23s\n- openstack-ansible-deploy-hosts_lxc_btrfs-ubuntu-noble https://zuul.opendev.org/t/openstack/build/9fe19ea5d515418990cac06e63f1b8a9 : SUCCESS in 24m 20s\n- openstack-ansible-deploy-hosts_lxc_overlayfs-ubuntu-noble https://zuul.opendev.org/t/openstack/build/ea5917c2a62f4c23ae4493d8cdaa728a : SUCCESS in 24m 41s\n- openstack-ansible-deploy-hosts_lxc_btrfs-debian-bookworm https://zuul.opendev.org/t/openstack/build/f14be18f177b4f35a310306dbb69b193 : SUCCESS in 25m 06s\n- openstack-ansible-deploy-hosts_lxc_overlayfs-debian-bookworm https://zuul.opendev.org/t/openstack/build/98a1611b59ac48c091a51059c779c370 : SUCCESS in 25m 46s\n- openstack-ansible-deploy-hosts_lxc_btrfs-debian-trixie https://zuul.opendev.org/t/openstack/build/6a5b393b54e4496fa2ccf6cb383ab019 : SUCCESS in 24m 01s\n- openstack-ansible-deploy-hosts_lxc_overlayfs-debian-trixie https://zuul.opendev.org/t/openstack/build/895d772e3df842dcbeb23f2bb3e71f1f : SUCCESS in 23m 54s\n- openstack-ansible-deploy-hosts_lxc_overlayfs-rockylinux-10 https://zuul.opendev.org/t/openstack/build/4c21fd5f17e44204bd6205c479c69806 : SUCCESS in 28m 34s","accounts_in_message":[],"_revision_number":4},{"id":"21e1615b39bb74f8b7c8f1813a9dd6d7983891a0","author":{"_account_id":37598,"name":"Ivan Anfimov","display_name":"Ivan Anfimov","email":"lazekteam@gmail.com","username":"anfimovir"},"date":"2026-05-25 06:40:52.000000000","message":"Patch Set 4:\n\n(1 comment)","accounts_in_message":[],"_revision_number":4},{"id":"4f5f25470509a40ccfd45e667fd70a408997977a","tag":"autogenerated:zuul:check","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2026-05-25 09:05:57.000000000","message":"Patch Set 4: Verified-1\n\nBuild failed (check pipeline).  For information on how to proceed, see\nhttps://docs.opendev.org/opendev/infra-manual/latest/developers.html#automated-testing\nand https://docs.openstack.org/project-team-guide/testing.html#how-to-handle-test-failures\n\nhttps://zuul.opendev.org/t/openstack/buildset/6af43dba9e824f5e9cf568b95e6d3796\n\n- openstack-ansible-linters-ubuntu-noble https://zuul.opendev.org/t/openstack/build/be5d904c96d34504aa15d6d7e847472a : SUCCESS in 5m 41s\n- openstack-tox-docs https://zuul.opendev.org/t/openstack/build/69d9c870849a45f3b4033069b122881c : SUCCESS in 6m 51s\n- build-openstack-releasenotes https://zuul.opendev.org/t/openstack/build/65de937814d54c2a88ec2f22cacb2800 : SUCCESS in 3m 05s\n- openstack-ansible-deploy-infra_lxc-centos-10-stream https://zuul.opendev.org/t/openstack/build/bea3d5f947224e6ba225d3b5106d2a55 : SUCCESS in 1h 49m 41s\n- openstack-ansible-deploy-infra_lxc-debian-bookworm https://zuul.opendev.org/t/openstack/build/dd415ab6ad87405f9ba13c64746885a4 : SUCCESS in 1h 55m 26s\n- openstack-ansible-deploy-infra_lxc-debian-trixie https://zuul.opendev.org/t/openstack/build/784d1cf2a117401d95ba4309b564c386 : SUCCESS in 1h 38m 06s\n- openstack-ansible-deploy-infra_lxc-rockylinux-10 https://zuul.opendev.org/t/openstack/build/8bcdccb82202454f816ceecf5e91287a : SUCCESS in 1h 39m 15s\n- openstack-ansible-deploy-infra_lxc_tls-rockylinux-10 https://zuul.opendev.org/t/openstack/build/3e3fbe919383494187df7ef5b6313cb7 : FAILURE in 58m 37s\n- openstack-ansible-deploy-infra_lxc-ubuntu-noble https://zuul.opendev.org/t/openstack/build/a0dc61b86c814ea39daabfd00b1b6120 : SUCCESS in 1h 29m 50s\n- openstack-ansible-deploy-infra_lxc_stepca-ubuntu-noble https://zuul.opendev.org/t/openstack/build/271adf0b65e542a5af534a4c467c2aa0 : SUCCESS in 1h 58m 50s\n- openstack-ansible-deploy-infra_lxc_tls-ubuntu-noble https://zuul.opendev.org/t/openstack/build/cbd5049d9f564d36805dd8ae4cf5f292 : SUCCESS in 2h 01m 46s\n- openstack-ansible-upgrade-infra_lxc-ubuntu-noble https://zuul.opendev.org/t/openstack/build/fc4dc55fba114e408942e72d07a0d10a : SUCCESS in 2h 14m 34s\n- openstack-ansible-deploy-hosts_lxc_zfs-ubuntu-noble https://zuul.opendev.org/t/openstack/build/68875356c84540908e01ad4693fefa8c : SUCCESS in 26m 44s\n- openstack-ansible-deploy-hosts_lxc_btrfs-ubuntu-noble https://zuul.opendev.org/t/openstack/build/c06ec8f9ce5b4cfe80b9ed4ff309d52a : SUCCESS in 16m 40s\n- openstack-ansible-deploy-hosts_lxc_overlayfs-ubuntu-noble https://zuul.opendev.org/t/openstack/build/4aa03e5d92594045a86b2d3614d45bbc : SUCCESS in 16m 26s\n- openstack-ansible-deploy-hosts_lxc_btrfs-debian-bookworm https://zuul.opendev.org/t/openstack/build/32377b9de33346c0a59ba89d3308bbd5 : SUCCESS in 27m 32s\n- openstack-ansible-deploy-hosts_lxc_overlayfs-debian-bookworm https://zuul.opendev.org/t/openstack/build/d045d0a2a4b74037905b471673f19432 : SUCCESS in 26m 48s\n- openstack-ansible-deploy-hosts_lxc_overlayfs-rockylinux-10 https://zuul.opendev.org/t/openstack/build/7258a12b0b4546a6b126802714720a63 : SUCCESS in 22m 41s\n- openstack-ansible-deploy-hosts_lxc_btrfs-debian-trixie https://zuul.opendev.org/t/openstack/build/47191f9838f84adfb0c1c564abd1eb35 : SUCCESS in 24m 28s\n- openstack-ansible-deploy-hosts_lxc_overlayfs-debian-trixie https://zuul.opendev.org/t/openstack/build/1eed576254b9480f8896ad61cca99893 : SUCCESS in 26m 16s","accounts_in_message":[],"_revision_number":4},{"id":"6804632178824b21f16c90482968fc2f843deaac","author":{"_account_id":28619,"name":"Dmitriy Rabotyagov","email":"noonedeadpunk@gmail.com","username":"noonedeadpunk"},"date":"2026-05-25 10:00:22.000000000","message":"Patch Set 4:\n\n(1 comment)","accounts_in_message":[],"_revision_number":4},{"id":"a5d1b07a6bc393c50e1b094a50cc9cff2787f3d4","tag":"autogenerated:zuul:check","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2026-05-25 12:41:15.000000000","message":"Patch Set 4: Verified+1\n\nBuild succeeded (check pipeline).\nhttps://zuul.opendev.org/t/openstack/buildset/78ffc9f891bc44549a0fa5df96333c01\n\n- openstack-ansible-linters-ubuntu-noble https://zuul.opendev.org/t/openstack/build/323cc5c243104cdda24c2b205bab8553 : SUCCESS in 12m 56s\n- openstack-tox-docs https://zuul.opendev.org/t/openstack/build/50e0437789fc4f5c9bcf630c127f8049 : SUCCESS in 7m 37s\n- build-openstack-releasenotes https://zuul.opendev.org/t/openstack/build/00114c1decd54066bf0d8c600b011174 : SUCCESS in 3m 38s\n- openstack-ansible-deploy-infra_lxc-centos-10-stream https://zuul.opendev.org/t/openstack/build/62ba52f961374c2f9f41d4cf5de9a3fb : SUCCESS in 2h 16m 45s\n- openstack-ansible-deploy-infra_lxc-debian-bookworm https://zuul.opendev.org/t/openstack/build/4c7519050e15471db7dad1e8fc24aaa5 : SUCCESS in 1h 52m 13s\n- openstack-ansible-deploy-infra_lxc-debian-trixie https://zuul.opendev.org/t/openstack/build/e0cd42c31a10423e81251fa51e8c3e64 : SUCCESS in 1h 55m 28s\n- openstack-ansible-deploy-infra_lxc-rockylinux-10 https://zuul.opendev.org/t/openstack/build/9d5abd33c5954a13a076a5473cb74af0 : SUCCESS in 1h 11m 49s\n- openstack-ansible-deploy-infra_lxc_tls-rockylinux-10 https://zuul.opendev.org/t/openstack/build/68a02f1f2d314acfacc59d675e1e3d4e : SUCCESS in 1h 44m 07s\n- openstack-ansible-deploy-infra_lxc-ubuntu-noble https://zuul.opendev.org/t/openstack/build/c42fb7027df4452ba23b8dc16831c0b2 : SUCCESS in 1h 49m 33s\n- openstack-ansible-deploy-infra_lxc_stepca-ubuntu-noble https://zuul.opendev.org/t/openstack/build/7253fba3182247cfa4d64278ea07efe7 : SUCCESS in 2h 31m 08s\n- openstack-ansible-deploy-infra_lxc_tls-ubuntu-noble https://zuul.opendev.org/t/openstack/build/442f7282c52c47fe83394bac998cc303 : SUCCESS in 1h 58m 26s\n- openstack-ansible-upgrade-infra_lxc-ubuntu-noble https://zuul.opendev.org/t/openstack/build/22adf69100dd4f09bc2ed4d7e03f626b : SUCCESS in 2h 16m 09s\n- openstack-ansible-deploy-hosts_lxc_zfs-ubuntu-noble https://zuul.opendev.org/t/openstack/build/0bc550e4eeb7478bbf8ad2d56369200e : SUCCESS in 26m 26s\n- openstack-ansible-deploy-hosts_lxc_btrfs-ubuntu-noble https://zuul.opendev.org/t/openstack/build/534280b35d79483b97cce9e5ba051fc0 : SUCCESS in 29m 27s\n- openstack-ansible-deploy-hosts_lxc_overlayfs-ubuntu-noble https://zuul.opendev.org/t/openstack/build/612d884992b440768d8302cec3ca1e71 : SUCCESS in 28m 41s\n- openstack-ansible-deploy-hosts_lxc_btrfs-debian-bookworm https://zuul.opendev.org/t/openstack/build/4a13faaf1bd84f2c94fe8d566a8eb351 : SUCCESS in 27m 59s\n- openstack-ansible-deploy-hosts_lxc_overlayfs-debian-bookworm https://zuul.opendev.org/t/openstack/build/f4958cef5c78484d9d4a0d3f62f41d8e : SUCCESS in 28m 59s\n- openstack-ansible-deploy-hosts_lxc_overlayfs-rockylinux-10 https://zuul.opendev.org/t/openstack/build/cd42664dcd4a42eda7cb555fae64533b : SUCCESS in 17m 30s\n- openstack-ansible-deploy-hosts_lxc_btrfs-debian-trixie https://zuul.opendev.org/t/openstack/build/16db1853e1aa4528b23eb9a30726b2ce : SUCCESS in 29m 02s\n- openstack-ansible-deploy-hosts_lxc_overlayfs-debian-trixie https://zuul.opendev.org/t/openstack/build/7f9566470eb3446a9ac92b36b249d9ed : SUCCESS in 29m 09s","accounts_in_message":[],"_revision_number":4},{"id":"dcfecb6cea6cccfbcc5a75b8c7a4b1effdcf89c7","tag":"autogenerated:zuul:gate","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2026-05-25 12:43:48.000000000","message":"Patch Set 4: -Verified\n\nStarting gate jobs.","accounts_in_message":[],"_revision_number":4},{"id":"8aebcbcc3e8c13341cd91056610c37b575533f8b","tag":"autogenerated:zuul:gate","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2026-05-25 14:50:16.000000000","message":"Patch Set 4: Verified+2\n\nBuild succeeded (gate pipeline).\nhttps://zuul.opendev.org/t/openstack/buildset/b163b5c8c8344d4388e81f083a460d56\n\n- openstack-ansible-linters-ubuntu-noble https://zuul.opendev.org/t/openstack/build/56aaaf2c95ee41f5a4566de7ccd2ccc6 : SUCCESS in 10m 06s\n- openstack-tox-docs https://zuul.opendev.org/t/openstack/build/96d8f182c4584dec97ce60b6b05ce715 : SUCCESS in 5m 55s\n- build-openstack-releasenotes https://zuul.opendev.org/t/openstack/build/c85b6238c9664f62bd71af36749b0018 : SUCCESS in 3m 21s\n- openstack-ansible-deploy-infra_lxc-centos-10-stream https://zuul.opendev.org/t/openstack/build/e45a92a002d748b6b97de7486d65b574 : SUCCESS in 1h 24m 58s\n- openstack-ansible-deploy-infra_lxc-debian-bookworm https://zuul.opendev.org/t/openstack/build/83e216b629f6489ea936fa9be01ea9e3 : SUCCESS in 1h 46m 44s\n- openstack-ansible-deploy-infra_lxc-debian-trixie https://zuul.opendev.org/t/openstack/build/2e8956b80eea4eb99f1f67036ff4bacf : SUCCESS in 1h 11m 05s\n- openstack-ansible-deploy-infra_lxc-rockylinux-10 https://zuul.opendev.org/t/openstack/build/0c0b5f9ffbdf4d5fb7087da5665bd541 : SUCCESS in 1h 32m 08s\n- openstack-ansible-deploy-infra_lxc_tls-rockylinux-10 https://zuul.opendev.org/t/openstack/build/bd8ba017f83b45ff963d8f0d1d3e3d8c : SUCCESS in 1h 16m 45s\n- openstack-ansible-deploy-infra_lxc-ubuntu-noble https://zuul.opendev.org/t/openstack/build/4901a650dd82427bad6143370cc7de89 : SUCCESS in 1h 45m 54s\n- openstack-ansible-deploy-infra_lxc_stepca-ubuntu-noble https://zuul.opendev.org/t/openstack/build/4aeadce6132d4f69b8b68792f6426c88 : SUCCESS in 1h 19m 16s\n- openstack-ansible-deploy-infra_lxc_tls-ubuntu-noble https://zuul.opendev.org/t/openstack/build/190d2456b5134d92b4b855b1c9513144 : SUCCESS in 1h 51m 25s\n- openstack-ansible-upgrade-infra_lxc-ubuntu-noble https://zuul.opendev.org/t/openstack/build/885652d72e4a48fe8bfd4c0a4c006701 : SUCCESS in 2h 06m 01s\n- openstack-ansible-deploy-hosts_lxc_zfs-ubuntu-noble https://zuul.opendev.org/t/openstack/build/c479eb26f6b5440597fd3c48ee5b511d : SUCCESS in 27m 24s\n- openstack-ansible-deploy-hosts_lxc_btrfs-ubuntu-noble https://zuul.opendev.org/t/openstack/build/70ece2fd99db460bba13203e6867be36 : SUCCESS in 23m 49s\n- openstack-ansible-deploy-hosts_lxc_overlayfs-ubuntu-noble https://zuul.opendev.org/t/openstack/build/a33f2486c2474cbdbff6650768e53bf7 : SUCCESS in 28m 09s\n- openstack-ansible-deploy-hosts_lxc_btrfs-debian-bookworm https://zuul.opendev.org/t/openstack/build/baf2bc960d5144d88d7409a3c137818e : SUCCESS in 23m 25s\n- openstack-ansible-deploy-hosts_lxc_overlayfs-debian-bookworm https://zuul.opendev.org/t/openstack/build/9d4325b9a05447f3aaae39cfd090162e : SUCCESS in 31m 55s\n- openstack-ansible-deploy-hosts_lxc_btrfs-debian-trixie https://zuul.opendev.org/t/openstack/build/9b151992a44b4fb79012576f951b061b : SUCCESS in 24m 30s\n- openstack-ansible-deploy-hosts_lxc_overlayfs-debian-trixie https://zuul.opendev.org/t/openstack/build/8aead853ea27457080fe086353c755a7 : SUCCESS in 25m 39s\n- openstack-ansible-deploy-hosts_lxc_overlayfs-rockylinux-10 https://zuul.opendev.org/t/openstack/build/2ce794e7d8c244f895278ce98d8392d7 : SUCCESS in 26m 52s","accounts_in_message":[],"_revision_number":4},{"id":"40dbdb022d67aa6ffa098e70c9f234913500e2dd","tag":"autogenerated:gerrit:merged","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2026-05-25 14:50:16.000000000","message":"Change has been successfully merged","accounts_in_message":[],"_revision_number":4},{"id":"2ad26c52f03ace535540283037a03ac62e1e4e59","tag":"autogenerated:zuul:promote","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2026-05-25 14:51:18.000000000","message":"Patch Set 4:\n\nBuild succeeded (promote pipeline).\nhttps://zuul.opendev.org/t/openstack/buildset/6be489282de445079fa0c697aa38c39a\n\n- promote-openstack-tox-docs https://zuul.opendev.org/t/openstack/build/a1fdae53a0f34be88345c7cd528efb15 : SUCCESS in 51s","accounts_in_message":[],"_revision_number":4}],"current_revision_number":4,"current_revision":"3d1b549fb553d665251107922b8dd2fb98787f94","revisions":{"f3cb0abd423b38c089a65163f9cc3f66865f1cbd":{"kind":"REWORK","_number":1,"created":"2026-04-18 09:10:32.000000000","uploader":{"_account_id":28619,"name":"Dmitriy Rabotyagov","email":"noonedeadpunk@gmail.com","username":"noonedeadpunk"},"ref":"refs/changes/16/985216/1","fetch":{"anonymous http":{"url":"https://review.opendev.org/openstack/openstack-ansible-lxc_hosts","ref":"refs/changes/16/985216/1","commands":{"Checkout":"git fetch https://review.opendev.org/openstack/openstack-ansible-lxc_hosts refs/changes/16/985216/1 \u0026\u0026 git checkout FETCH_HEAD","Cherry Pick":"git fetch https://review.opendev.org/openstack/openstack-ansible-lxc_hosts refs/changes/16/985216/1 \u0026\u0026 git cherry-pick FETCH_HEAD","Format Patch":"git fetch https://review.opendev.org/openstack/openstack-ansible-lxc_hosts refs/changes/16/985216/1 \u0026\u0026 git format-patch -1 --stdout FETCH_HEAD","Pull":"git pull https://review.opendev.org/openstack/openstack-ansible-lxc_hosts refs/changes/16/985216/1"}}},"commit":{"parents":[{"commit":"c440331b9bb49e527977a9d3eba902cb40a480a2","subject":"Merge \"Remove copying a non-existent configuration file in LXC\"","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/openstack-ansible-lxc_hosts/commit/c440331b9bb49e527977a9d3eba902cb40a480a2"}]}],"author":{"name":"Dmitriy Rabotyagov","email":"dmitriy.rabotyagov@cleura.com","date":"2026-04-18 09:10:25.000000000","tz":120},"committer":{"name":"Dmitriy Rabotyagov","email":"dmitriy.rabotyagov@cleura.com","date":"2026-04-18 09:10:25.000000000","tz":120},"subject":"Add support for Ubuntu 26.04","message":"Add support for Ubuntu 26.04\n\nChange-Id: I2ceb1040e978dfe4fcd552dfd67d68050f552b3c\nSigned-off-by: Dmitriy Rabotyagov \u003cdmitriy.rabotyagov@cleura.com\u003e\n","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/openstack-ansible-lxc_hosts/commit/f3cb0abd423b38c089a65163f9cc3f66865f1cbd"}],"resolve_conflicts_web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/openstack-ansible-lxc_hosts/commit/f3cb0abd423b38c089a65163f9cc3f66865f1cbd"}]},"branch":"refs/heads/master"},"664c6fa3ee01b156dc350006b37e08cfe766f0c5":{"kind":"REWORK","_number":2,"created":"2026-04-18 09:45:58.000000000","uploader":{"_account_id":28619,"name":"Dmitriy Rabotyagov","email":"noonedeadpunk@gmail.com","username":"noonedeadpunk"},"ref":"refs/changes/16/985216/2","fetch":{"anonymous http":{"url":"https://review.opendev.org/openstack/openstack-ansible-lxc_hosts","ref":"refs/changes/16/985216/2","commands":{"Checkout":"git fetch https://review.opendev.org/openstack/openstack-ansible-lxc_hosts refs/changes/16/985216/2 \u0026\u0026 git checkout FETCH_HEAD","Cherry Pick":"git fetch https://review.opendev.org/openstack/openstack-ansible-lxc_hosts refs/changes/16/985216/2 \u0026\u0026 git cherry-pick FETCH_HEAD","Format Patch":"git fetch https://review.opendev.org/openstack/openstack-ansible-lxc_hosts refs/changes/16/985216/2 \u0026\u0026 git format-patch -1 --stdout FETCH_HEAD","Pull":"git pull https://review.opendev.org/openstack/openstack-ansible-lxc_hosts refs/changes/16/985216/2"}}},"commit":{"parents":[{"commit":"c440331b9bb49e527977a9d3eba902cb40a480a2","subject":"Merge \"Remove copying a non-existent configuration file in LXC\"","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/openstack-ansible-lxc_hosts/commit/c440331b9bb49e527977a9d3eba902cb40a480a2"}]}],"author":{"name":"Dmitriy Rabotyagov","email":"dmitriy.rabotyagov@cleura.com","date":"2026-04-18 09:10:25.000000000","tz":120},"committer":{"name":"Dmitriy Rabotyagov","email":"dmitriy.rabotyagov@cleura.com","date":"2026-04-18 09:45:52.000000000","tz":120},"subject":"Add support for Ubuntu 26.04","message":"Add support for Ubuntu 26.04\n\nChange-Id: I2ceb1040e978dfe4fcd552dfd67d68050f552b3c\nSigned-off-by: Dmitriy Rabotyagov \u003cdmitriy.rabotyagov@cleura.com\u003e\n","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/openstack-ansible-lxc_hosts/commit/664c6fa3ee01b156dc350006b37e08cfe766f0c5"}],"resolve_conflicts_web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/openstack-ansible-lxc_hosts/commit/664c6fa3ee01b156dc350006b37e08cfe766f0c5"}]},"branch":"refs/heads/master"},"d55b8a0c41dedbf0f1021caa40b91e13e2cb387b":{"kind":"REWORK","_number":3,"created":"2026-04-18 09:57:51.000000000","uploader":{"_account_id":28619,"name":"Dmitriy Rabotyagov","email":"noonedeadpunk@gmail.com","username":"noonedeadpunk"},"ref":"refs/changes/16/985216/3","fetch":{"anonymous http":{"url":"https://review.opendev.org/openstack/openstack-ansible-lxc_hosts","ref":"refs/changes/16/985216/3","commands":{"Checkout":"git fetch https://review.opendev.org/openstack/openstack-ansible-lxc_hosts refs/changes/16/985216/3 \u0026\u0026 git checkout FETCH_HEAD","Cherry Pick":"git fetch https://review.opendev.org/openstack/openstack-ansible-lxc_hosts refs/changes/16/985216/3 \u0026\u0026 git cherry-pick FETCH_HEAD","Format Patch":"git fetch https://review.opendev.org/openstack/openstack-ansible-lxc_hosts refs/changes/16/985216/3 \u0026\u0026 git format-patch -1 --stdout FETCH_HEAD","Pull":"git pull https://review.opendev.org/openstack/openstack-ansible-lxc_hosts refs/changes/16/985216/3"}}},"commit":{"parents":[{"commit":"c440331b9bb49e527977a9d3eba902cb40a480a2","subject":"Merge \"Remove copying a non-existent configuration file in LXC\"","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/openstack-ansible-lxc_hosts/commit/c440331b9bb49e527977a9d3eba902cb40a480a2"}]}],"author":{"name":"Dmitriy Rabotyagov","email":"dmitriy.rabotyagov@cleura.com","date":"2026-04-18 09:10:25.000000000","tz":120},"committer":{"name":"Dmitriy Rabotyagov","email":"dmitriy.rabotyagov@cleura.com","date":"2026-04-18 09:57:16.000000000","tz":120},"subject":"Add support for Ubuntu 26.04","message":"Add support for Ubuntu 26.04\n\nWe also clean-up logic for currently unsupported Ubuntu/Debian distributions\n\nChange-Id: I2ceb1040e978dfe4fcd552dfd67d68050f552b3c\nSigned-off-by: Dmitriy Rabotyagov \u003cdmitriy.rabotyagov@cleura.com\u003e\n","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/openstack-ansible-lxc_hosts/commit/d55b8a0c41dedbf0f1021caa40b91e13e2cb387b"}],"resolve_conflicts_web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/openstack-ansible-lxc_hosts/commit/d55b8a0c41dedbf0f1021caa40b91e13e2cb387b"}]},"branch":"refs/heads/master"},"3d1b549fb553d665251107922b8dd2fb98787f94":{"kind":"REWORK","_number":4,"created":"2026-04-21 08:35:22.000000000","uploader":{"_account_id":28619,"name":"Dmitriy Rabotyagov","email":"noonedeadpunk@gmail.com","username":"noonedeadpunk"},"ref":"refs/changes/16/985216/4","fetch":{"anonymous http":{"url":"https://review.opendev.org/openstack/openstack-ansible-lxc_hosts","ref":"refs/changes/16/985216/4","commands":{"Checkout":"git fetch https://review.opendev.org/openstack/openstack-ansible-lxc_hosts refs/changes/16/985216/4 \u0026\u0026 git checkout FETCH_HEAD","Cherry Pick":"git fetch https://review.opendev.org/openstack/openstack-ansible-lxc_hosts refs/changes/16/985216/4 \u0026\u0026 git cherry-pick FETCH_HEAD","Format Patch":"git fetch https://review.opendev.org/openstack/openstack-ansible-lxc_hosts refs/changes/16/985216/4 \u0026\u0026 git format-patch -1 --stdout FETCH_HEAD","Pull":"git pull https://review.opendev.org/openstack/openstack-ansible-lxc_hosts refs/changes/16/985216/4"}}},"commit":{"parents":[{"commit":"c440331b9bb49e527977a9d3eba902cb40a480a2","subject":"Merge \"Remove copying a non-existent configuration file in LXC\"","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/openstack-ansible-lxc_hosts/commit/c440331b9bb49e527977a9d3eba902cb40a480a2"}]}],"author":{"name":"Dmitriy Rabotyagov","email":"dmitriy.rabotyagov@cleura.com","date":"2026-04-18 09:10:25.000000000","tz":120},"committer":{"name":"Dmitriy Rabotyagov","email":"dmitriy.rabotyagov@cleura.com","date":"2026-04-21 08:35:22.000000000","tz":0},"subject":"Add support for Ubuntu 26.04","message":"Add support for Ubuntu 26.04\n\nWe also clean-up logic for currently unsupported Ubuntu/Debian distributions\n\nChange-Id: I2ceb1040e978dfe4fcd552dfd67d68050f552b3c\nSigned-off-by: Dmitriy Rabotyagov \u003cdmitriy.rabotyagov@cleura.com\u003e\n","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/openstack-ansible-lxc_hosts/commit/3d1b549fb553d665251107922b8dd2fb98787f94"}],"resolve_conflicts_web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/openstack-ansible-lxc_hosts/commit/3d1b549fb553d665251107922b8dd2fb98787f94"}]},"branch":"refs/heads/master"}},"requirements":[],"submit_records":[{"rule_name":"gerrit~DefaultSubmitRule","status":"CLOSED","labels":[{"label":"Verified","status":"MAY","applied_by":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]}},{"label":"Code-Review","status":"MAY","applied_by":{"_account_id":25023,"name":"Jonathan Rosser","email":"jonathan.rosser@rd.bbc.co.uk","username":"jrosser"}},{"label":"Workflow","status":"MAY","applied_by":{"_account_id":32666,"name":"Damian Dąbrowski","email":"damian@dabrowski.cloud","username":"ddabrowski"}},{"label":"Backport-Candidate","status":"MAY"}]}],"submit_requirements":[{"name":"Verified","description":"Verified in gate by CI","status":"SATISFIED","is_legacy":false,"submittability_expression_result":{"expression":"label:Verified\u003dMAX AND -label:Verified\u003dMIN","fulfilled":true,"status":"PASS","passing_atoms":["label:Verified\u003dMAX"],"failing_atoms":["label:Verified\u003dMIN"],"atom_explanations":{"label:Verified\u003dMAX":"","label:Verified\u003dMIN":""}}},{"name":"Backport-Candidate","description":"Backport candidate status","status":"NOT_APPLICABLE","is_legacy":false,"applicability_expression_result":{"fulfilled":false,"status":"FAIL"},"submittability_expression_result":{"expression":"is:true","fulfilled":true,"status":"NOT_EVALUATED","passing_atoms":[],"failing_atoms":[],"atom_explanations":{}}},{"name":"Code-Review","description":"Code reviewed by core reviewer","status":"SATISFIED","is_legacy":false,"submittability_expression_result":{"expression":"label:Code-Review\u003dMAX AND -label:Code-Review\u003dMIN","fulfilled":true,"status":"PASS","passing_atoms":["label:Code-Review\u003dMAX"],"failing_atoms":["label:Code-Review\u003dMIN"],"atom_explanations":{"label:Code-Review\u003dMAX":"","label:Code-Review\u003dMIN":""}}},{"name":"Workflow","description":"Approved for gate by core reviewer","status":"SATISFIED","is_legacy":false,"submittability_expression_result":{"expression":"label:Workflow\u003dMAX AND -label:Workflow\u003dMIN","fulfilled":true,"status":"PASS","passing_atoms":["label:Workflow\u003dMAX"],"failing_atoms":["label:Workflow\u003dMIN"],"atom_explanations":{"label:Workflow\u003dMAX":"","label:Workflow\u003dMIN":""}}}]}
