)]}'
{"id":"openstack%2Fopenstack-ansible-lxc_hosts~990070","triplet_id":"openstack%2Fopenstack-ansible-lxc_hosts~master~If5f328417ca0c2ec427de70181dd5623c1802029","project":"openstack/openstack-ansible-lxc_hosts","branch":"master","topic":"osa/resolute","attention_set":{},"removed_from_attention_set":{"31542":{"account":{"_account_id":31542,"name":"Andrew Bonney","email":"andrew.bonney@bbc.co.uk","username":"andrewbonney"},"last_update":"2026-06-02 12:46:57.000000000","reason":"\u003cGERRIT_ACCOUNT_31542\u003e replied on the change","reason_account":{"_account_id":31542,"name":"Andrew Bonney","email":"andrew.bonney@bbc.co.uk","username":"andrewbonney"}},"28619":{"account":{"_account_id":28619,"name":"Dmitriy Rabotyagov","email":"noonedeadpunk@gmail.com","username":"noonedeadpunk"},"last_update":"2026-06-02 15:03:11.000000000","reason":"\u003cGERRIT_ACCOUNT_28619\u003e replied on the change","reason_account":{"_account_id":28619,"name":"Dmitriy Rabotyagov","email":"noonedeadpunk@gmail.com","username":"noonedeadpunk"}},"25023":{"account":{"_account_id":25023,"name":"Jonathan Rosser","email":"jonathan.rosser@rd.bbc.co.uk","username":"jrosser"},"last_update":"2026-06-02 17:31:34.000000000","reason":"Change was submitted"}},"hashtags":[],"change_id":"If5f328417ca0c2ec427de70181dd5623c1802029","subject":"Ensure ubuntu_virt hook not passed to LXC","status":"MERGED","created":"2026-05-26 09:13:15.000000000","updated":"2026-06-02 17:33:13.000000000","submitted":"2026-06-02 17:31:34.000000000","submitter":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"total_comment_count":7,"unresolved_comment_count":0,"has_review_started":true,"submission_id":"990070-osa/resolute","meta_rev_id":"7ae33803825b5fdc9b9bf7b322cad03d7edc2702","_number":990070,"virtual_id_number":990070,"owner":{"_account_id":28619,"name":"Dmitriy Rabotyagov","email":"noonedeadpunk@gmail.com","username":"noonedeadpunk"},"actions":{},"labels":{"Verified":{"approved":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"all":[{"value":0,"_account_id":31542,"name":"Andrew Bonney","email":"andrew.bonney@bbc.co.uk","username":"andrewbonney"},{"value":0,"_account_id":28619,"name":"Dmitriy Rabotyagov","email":"noonedeadpunk@gmail.com","username":"noonedeadpunk"},{"tag":"autogenerated:zuul:gate","value":2,"date":"2026-06-02 17:31:34.000000000","permitted_voting_range":{"min":2,"max":2},"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},{"value":0,"_account_id":37598,"name":"Ivan Anfimov","display_name":"Ivan Anfimov","email":"lazekteam@gmail.com","username":"anfimovir"},{"value":0,"_account_id":25023,"name":"Jonathan Rosser","email":"jonathan.rosser@rd.bbc.co.uk","username":"jrosser"}],"values":{"-2":"Fails","-1":"Doesn\u0027t seem to work"," 0":"No score","+1":"Works for me","+2":"Verified"},"description":"","default_value":0,"optional":true},"Code-Review":{"approved":{"_account_id":31542,"name":"Andrew Bonney","email":"andrew.bonney@bbc.co.uk","username":"andrewbonney"},"all":[{"value":2,"date":"2026-06-02 12:46:57.000000000","permitted_voting_range":{"min":2,"max":2},"_account_id":31542,"name":"Andrew Bonney","email":"andrew.bonney@bbc.co.uk","username":"andrewbonney"},{"value":0,"_account_id":28619,"name":"Dmitriy Rabotyagov","email":"noonedeadpunk@gmail.com","username":"noonedeadpunk"},{"value":0,"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},{"value":2,"date":"2026-06-02 12:51:53.000000000","permitted_voting_range":{"min":2,"max":2},"_account_id":37598,"name":"Ivan Anfimov","display_name":"Ivan Anfimov","email":"lazekteam@gmail.com","username":"anfimovir"},{"value":0,"_account_id":25023,"name":"Jonathan Rosser","email":"jonathan.rosser@rd.bbc.co.uk","username":"jrosser"}],"values":{"-2":"Do not merge","-1":"This patch needs further work before it can be merged"," 0":"No score","+1":"Looks good to me, but someone else must approve","+2":"Looks good to me (core reviewer)"},"description":"","default_value":0,"optional":true},"Workflow":{"approved":{"_account_id":28619,"name":"Dmitriy Rabotyagov","email":"noonedeadpunk@gmail.com","username":"noonedeadpunk"},"all":[{"value":0,"_account_id":31542,"name":"Andrew Bonney","email":"andrew.bonney@bbc.co.uk","username":"andrewbonney"},{"value":1,"date":"2026-06-02 15:03:11.000000000","permitted_voting_range":{"min":1,"max":1},"_account_id":28619,"name":"Dmitriy Rabotyagov","email":"noonedeadpunk@gmail.com","username":"noonedeadpunk"},{"value":0,"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},{"value":0,"_account_id":37598,"name":"Ivan Anfimov","display_name":"Ivan Anfimov","email":"lazekteam@gmail.com","username":"anfimovir"},{"value":0,"_account_id":25023,"name":"Jonathan Rosser","email":"jonathan.rosser@rd.bbc.co.uk","username":"jrosser"}],"values":{"-1":"Work in progress"," 0":"Ready for reviews","+1":"Approved"},"description":"","default_value":0,"optional":true},"Backport-Candidate":{"all":[{"value":0,"_account_id":31542,"name":"Andrew Bonney","email":"andrew.bonney@bbc.co.uk","username":"andrewbonney"},{"value":0,"_account_id":28619,"name":"Dmitriy Rabotyagov","email":"noonedeadpunk@gmail.com","username":"noonedeadpunk"},{"value":0,"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},{"value":0,"_account_id":37598,"name":"Ivan Anfimov","display_name":"Ivan Anfimov","email":"lazekteam@gmail.com","username":"anfimovir"},{"value":0,"_account_id":25023,"name":"Jonathan Rosser","email":"jonathan.rosser@rd.bbc.co.uk","username":"jrosser"}],"values":{"-1":"Not A Backport Candidate"," 0":"Backport Review Needed","+1":"Proposed Backport"},"description":"","default_value":0,"optional":true}},"removable_reviewers":[],"reviewers":{"REVIEWER":[{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},{"_account_id":25023,"name":"Jonathan Rosser","email":"jonathan.rosser@rd.bbc.co.uk","username":"jrosser"},{"_account_id":28619,"name":"Dmitriy Rabotyagov","email":"noonedeadpunk@gmail.com","username":"noonedeadpunk"},{"_account_id":31542,"name":"Andrew Bonney","email":"andrew.bonney@bbc.co.uk","username":"andrewbonney"},{"_account_id":37598,"name":"Ivan Anfimov","display_name":"Ivan Anfimov","email":"lazekteam@gmail.com","username":"anfimovir"}]},"pending_reviewers":{},"reviewer_updates":[{"updated":"2026-05-26 11:46:20.000000000","updated_by":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"reviewer":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"state":"REVIEWER"},{"updated":"2026-05-27 08:37:43.000000000","updated_by":{"_account_id":31542,"name":"Andrew Bonney","email":"andrew.bonney@bbc.co.uk","username":"andrewbonney"},"reviewer":{"_account_id":31542,"name":"Andrew Bonney","email":"andrew.bonney@bbc.co.uk","username":"andrewbonney"},"state":"REVIEWER"},{"updated":"2026-05-28 17:18:42.000000000","updated_by":{"_account_id":25023,"name":"Jonathan Rosser","email":"jonathan.rosser@rd.bbc.co.uk","username":"jrosser"},"reviewer":{"_account_id":25023,"name":"Jonathan Rosser","email":"jonathan.rosser@rd.bbc.co.uk","username":"jrosser"},"state":"REVIEWER"},{"updated":"2026-05-30 20:26:08.000000000","updated_by":{"_account_id":37598,"name":"Ivan Anfimov","display_name":"Ivan Anfimov","email":"lazekteam@gmail.com","username":"anfimovir"},"reviewer":{"_account_id":37598,"name":"Ivan Anfimov","display_name":"Ivan Anfimov","email":"lazekteam@gmail.com","username":"anfimovir"},"state":"CC"},{"updated":"2026-06-02 12:51:53.000000000","updated_by":{"_account_id":37598,"name":"Ivan Anfimov","display_name":"Ivan Anfimov","email":"lazekteam@gmail.com","username":"anfimovir"},"reviewer":{"_account_id":37598,"name":"Ivan Anfimov","display_name":"Ivan Anfimov","email":"lazekteam@gmail.com","username":"anfimovir"},"state":"REVIEWER"}],"messages":[{"id":"1a211f5bddc3f352176718c4f7bcfbef524691c7","tag":"autogenerated:gerrit:newPatchSet","author":{"_account_id":28619,"name":"Dmitriy Rabotyagov","email":"noonedeadpunk@gmail.com","username":"noonedeadpunk"},"date":"2026-05-26 09:13:15.000000000","message":"Uploaded patch set 1.","accounts_in_message":[],"_revision_number":1},{"id":"9ea001f973f4dcea4f7717afbb21e2d8e05a9b27","tag":"autogenerated:zuul:check","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2026-05-26 11:46:20.000000000","message":"Patch Set 1: Verified-1\n\nBuild failed (check pipeline).  For information on how to proceed, see\nhttps://docs.opendev.org/opendev/infra-manual/latest/developers.html#automated-testing\nand https://docs.openstack.org/project-team-guide/testing.html#how-to-handle-test-failures\n\nhttps://zuul.opendev.org/t/openstack/buildset/178a5375808645e796837baf40b1fb0f\n\n- openstack-ansible-linters-ubuntu-noble https://zuul.opendev.org/t/openstack/build/371f7f3614b7407a9a773e02980e08db : SUCCESS in 16m 02s\n- openstack-tox-docs https://zuul.opendev.org/t/openstack/build/b78f844926d0434394b01332278fd604 : SUCCESS in 6m 42s\n- build-openstack-releasenotes https://zuul.opendev.org/t/openstack/build/4c9b2117a32d4ed8b9e55125602f0ee9 : SUCCESS in 3m 16s\n- openstack-ansible-deploy-infra_lxc-centos-10-stream https://zuul.opendev.org/t/openstack/build/3cb39964d5584011a4881176940ba918 : SUCCESS in 2h 24m 20s\n- openstack-ansible-deploy-infra_lxc-debian-bookworm https://zuul.opendev.org/t/openstack/build/499352cbde78495699062fab40f8f065 : SUCCESS in 1h 42m 09s\n- openstack-ansible-deploy-infra_lxc-debian-trixie https://zuul.opendev.org/t/openstack/build/1aa2e372f9074d858a48e7c898cc16fb : SUCCESS in 1h 58m 48s\n- openstack-ansible-deploy-infra_lxc-rockylinux-10 https://zuul.opendev.org/t/openstack/build/6ffac20c92aa40df83f714bcfd1482bc : FAILURE in 2h 17m 11s\n- openstack-ansible-deploy-infra_lxc_tls-rockylinux-10 https://zuul.opendev.org/t/openstack/build/ba9f4daf8c024c8f84d040b41928fea4 : SUCCESS in 2h 02m 06s\n- openstack-ansible-deploy-infra_lxc-ubuntu-noble https://zuul.opendev.org/t/openstack/build/57f9d4f008bb4f23b04b0a7d45d68ed4 : SUCCESS in 2h 07m 09s\n- openstack-ansible-deploy-infra_lxc_stepca-ubuntu-noble https://zuul.opendev.org/t/openstack/build/6d36280fa4784bad98c407c62bc2414b : SUCCESS in 1h 52m 26s\n- openstack-ansible-deploy-infra_lxc_tls-ubuntu-noble https://zuul.opendev.org/t/openstack/build/888c897b1616417680a1db0923c5bbb7 : SUCCESS in 2h 01m 10s\n- openstack-ansible-upgrade-infra_lxc-ubuntu-noble https://zuul.opendev.org/t/openstack/build/efda87d8226045b8bf7cf10e7bfb08f9 : SUCCESS in 2h 22m 30s\n- openstack-ansible-deploy-hosts_lxc_zfs-ubuntu-noble https://zuul.opendev.org/t/openstack/build/fa320ab90e094c94b791696168884030 : SUCCESS in 26m 18s\n- openstack-ansible-deploy-hosts_lxc_btrfs-ubuntu-noble https://zuul.opendev.org/t/openstack/build/706bcad5884a4b28b0e3aabbf1c9b4c3 : SUCCESS in 32m 56s\n- openstack-ansible-deploy-hosts_lxc_overlayfs-ubuntu-noble https://zuul.opendev.org/t/openstack/build/ad3020808daf4920a18286419284332b : SUCCESS in 29m 23s\n- openstack-ansible-deploy-hosts_lxc_btrfs-debian-bookworm https://zuul.opendev.org/t/openstack/build/151877269f914dfd882838240025938e : SUCCESS in 33m 54s\n- openstack-ansible-deploy-hosts_lxc_overlayfs-debian-bookworm https://zuul.opendev.org/t/openstack/build/2b4b10cd5d984108a43fb7968fc7c2e9 : SUCCESS in 31m 24s\n- openstack-ansible-deploy-hosts_lxc_overlayfs-rockylinux-10 https://zuul.opendev.org/t/openstack/build/f4fb011b86824a4c9b63d13007e3e17c : SUCCESS in 28m 56s\n- openstack-ansible-deploy-hosts_lxc_btrfs-debian-trixie https://zuul.opendev.org/t/openstack/build/ef8524b519554148ae5894aee3f9c8d1 : SUCCESS in 29m 19s\n- openstack-ansible-deploy-hosts_lxc_overlayfs-debian-trixie https://zuul.opendev.org/t/openstack/build/98bbaee61d0a41978db9b011c75c65e7 : SUCCESS in 29m 03s","accounts_in_message":[],"_revision_number":1},{"id":"97efcf78077ecd40addab7996475a617a8f47523","author":{"_account_id":28619,"name":"Dmitriy Rabotyagov","email":"noonedeadpunk@gmail.com","username":"noonedeadpunk"},"date":"2026-05-26 15:27:28.000000000","message":"Patch Set 1:\n\n(1 comment)","accounts_in_message":[],"_revision_number":1},{"id":"94be131f1330e65da52091e98c7718aaeb2443ea","tag":"autogenerated:zuul:check","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2026-05-26 17:49:01.000000000","message":"Patch Set 1:\n\nBuild failed (check pipeline).  For information on how to proceed, see\nhttps://docs.opendev.org/opendev/infra-manual/latest/developers.html#automated-testing\nand https://docs.openstack.org/project-team-guide/testing.html#how-to-handle-test-failures\n\nhttps://zuul.opendev.org/t/openstack/buildset/c03e57345cf645d792986de7aeacd747\n\n- openstack-ansible-linters-ubuntu-noble https://zuul.opendev.org/t/openstack/build/92d7a7411e7a4f189c533681e95606f0 : SUCCESS in 9m 56s\n- openstack-tox-docs https://zuul.opendev.org/t/openstack/build/5b4523b0b41b4891a0fefd4bdeb315b9 : SUCCESS in 5m 57s\n- build-openstack-releasenotes https://zuul.opendev.org/t/openstack/build/4a849a9f48fb4effbf0e4b56e395c3ad : SUCCESS in 4m 41s\n- openstack-ansible-deploy-infra_lxc-centos-10-stream https://zuul.opendev.org/t/openstack/build/cfb1f38ee467447fabaa3968e3586175 : FAILURE in 2h 06m 38s\n- openstack-ansible-deploy-infra_lxc-debian-bookworm https://zuul.opendev.org/t/openstack/build/63d15fdf85f1434981b35ba028a80b72 : SUCCESS in 1h 51m 42s\n- openstack-ansible-deploy-infra_lxc-debian-trixie https://zuul.opendev.org/t/openstack/build/161dae099a5841c0a61f825fea2a2358 : SUCCESS in 1h 22m 41s\n- openstack-ansible-deploy-infra_lxc-rockylinux-10 https://zuul.opendev.org/t/openstack/build/7efbb7d74fce45f2be7b60bd697147cb : SUCCESS in 1h 21m 13s\n- openstack-ansible-deploy-infra_lxc_tls-rockylinux-10 https://zuul.opendev.org/t/openstack/build/3fc640e6931044778cbdf9fc90cd3087 : SUCCESS in 2h 06m 07s\n- openstack-ansible-deploy-infra_lxc-ubuntu-noble https://zuul.opendev.org/t/openstack/build/1ca78b379e6c421f9e2e204ae6bbf4ae : SUCCESS in 1h 52m 44s\n- openstack-ansible-deploy-infra_lxc_stepca-ubuntu-noble https://zuul.opendev.org/t/openstack/build/a50fc35005c2487e871687abdf05b3a9 : SUCCESS in 1h 58m 24s\n- openstack-ansible-deploy-infra_lxc_tls-ubuntu-noble https://zuul.opendev.org/t/openstack/build/7286cb704f494e33a28eca235015d7f3 : SUCCESS in 1h 57m 11s\n- openstack-ansible-upgrade-infra_lxc-ubuntu-noble https://zuul.opendev.org/t/openstack/build/01fc6d6676794b2cbd1a5b352818599e : SUCCESS in 2h 18m 08s\n- openstack-ansible-deploy-hosts_lxc_zfs-ubuntu-noble https://zuul.opendev.org/t/openstack/build/c47b1564403e4a368eba984f46310f68 : SUCCESS in 19m 22s\n- openstack-ansible-deploy-hosts_lxc_btrfs-ubuntu-noble https://zuul.opendev.org/t/openstack/build/5441d35ee044411ca9ac3ea5dacb22a4 : SUCCESS in 17m 16s\n- openstack-ansible-deploy-hosts_lxc_overlayfs-ubuntu-noble https://zuul.opendev.org/t/openstack/build/3d31689429a04997875b17b5a8e24f02 : SUCCESS in 26m 36s\n- openstack-ansible-deploy-hosts_lxc_btrfs-debian-bookworm https://zuul.opendev.org/t/openstack/build/18010aa5bc7048dca64151842323688a : SUCCESS in 16m 52s\n- openstack-ansible-deploy-hosts_lxc_overlayfs-debian-bookworm https://zuul.opendev.org/t/openstack/build/858f59c3b5914dba86bd65323c98edab : SUCCESS in 30m 00s\n- openstack-ansible-deploy-hosts_lxc_overlayfs-rockylinux-10 https://zuul.opendev.org/t/openstack/build/bfa18d7387684eff8b063a7d6fba31f8 : SUCCESS in 31m 46s\n- openstack-ansible-deploy-hosts_lxc_btrfs-debian-trixie https://zuul.opendev.org/t/openstack/build/a98f6e8f66714fb8bc12f911af6151a6 : SUCCESS in 25m 28s\n- openstack-ansible-deploy-hosts_lxc_overlayfs-debian-trixie https://zuul.opendev.org/t/openstack/build/802a1e2c746f4a098c6b662cfad52dcf : SUCCESS in 26m 48s","accounts_in_message":[],"_revision_number":1},{"id":"83d85edcf13279251d0c6d649427c60163b06062","author":{"_account_id":28619,"name":"Dmitriy Rabotyagov","email":"noonedeadpunk@gmail.com","username":"noonedeadpunk"},"date":"2026-05-26 18:20:13.000000000","message":"Patch Set 1:\n\n(1 comment)","accounts_in_message":[],"_revision_number":1},{"id":"fb3179b5b2c42931c43490bd799cae555821bf9c","tag":"autogenerated:zuul:check","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2026-05-26 21:23:19.000000000","message":"Patch Set 1: Verified+1\n\nBuild succeeded (check pipeline).\nhttps://zuul.opendev.org/t/openstack/buildset/2bda9b96a86b47e69e6f5d6467d23591\n\n- openstack-ansible-linters-ubuntu-noble https://zuul.opendev.org/t/openstack/build/8fb16ee833db4e80a439c7866b242151 : SUCCESS in 12m 31s\n- openstack-tox-docs https://zuul.opendev.org/t/openstack/build/07f16cf6f4d74d96ac51595d45578ffb : SUCCESS in 7m 37s\n- build-openstack-releasenotes https://zuul.opendev.org/t/openstack/build/02466c7101cc48ababd79d2e8231e5c1 : SUCCESS in 4m 12s\n- openstack-ansible-deploy-infra_lxc-centos-10-stream https://zuul.opendev.org/t/openstack/build/c0e4dce32e88425ca7bf7acc3fc414d8 : SUCCESS in 1h 22m 18s\n- openstack-ansible-deploy-infra_lxc-debian-bookworm https://zuul.opendev.org/t/openstack/build/65185b72fa534e94a8347f5a7da21527 : SUCCESS in 1h 50m 13s\n- openstack-ansible-deploy-infra_lxc-debian-trixie https://zuul.opendev.org/t/openstack/build/6aecfe5550eb44a8a166fb42f65eab6b : SUCCESS in 1h 45m 47s\n- openstack-ansible-deploy-infra_lxc-rockylinux-10 https://zuul.opendev.org/t/openstack/build/84cf673dd7f248cd9c745eed42870813 : SUCCESS in 2h 09m 48s\n- openstack-ansible-deploy-infra_lxc_tls-rockylinux-10 https://zuul.opendev.org/t/openstack/build/7f3372b30c0d450ebcd7c414ccdd4f2f : SUCCESS in 2h 11m 03s\n- openstack-ansible-deploy-infra_lxc-ubuntu-noble https://zuul.opendev.org/t/openstack/build/3ab9a76484724689bc4aaf2eb13103fd : SUCCESS in 1h 45m 25s\n- openstack-ansible-deploy-infra_lxc_stepca-ubuntu-noble https://zuul.opendev.org/t/openstack/build/aa6f442481fc4c988f7567a90a51efe6 : SUCCESS in 1h 55m 14s\n- openstack-ansible-deploy-infra_lxc_tls-ubuntu-noble https://zuul.opendev.org/t/openstack/build/1007f41ab6d4442eab72b12e19ca9d08 : SUCCESS in 1h 45m 15s\n- openstack-ansible-upgrade-infra_lxc-ubuntu-noble https://zuul.opendev.org/t/openstack/build/3b0da14e7cc544be92c5b6f44cc3fd88 : SUCCESS in 2h 28m 53s\n- openstack-ansible-deploy-hosts_lxc_zfs-ubuntu-noble https://zuul.opendev.org/t/openstack/build/a7a5b71f3e394ef98162a769d7efa555 : SUCCESS in 31m 39s\n- openstack-ansible-deploy-hosts_lxc_btrfs-ubuntu-noble https://zuul.opendev.org/t/openstack/build/29d75e8d358a40c2b59f3d8154c45d69 : SUCCESS in 30m 41s\n- openstack-ansible-deploy-hosts_lxc_overlayfs-ubuntu-noble https://zuul.opendev.org/t/openstack/build/015f2bcd59a1473fa4a091ef77701adf : SUCCESS in 29m 36s\n- openstack-ansible-deploy-hosts_lxc_btrfs-debian-bookworm https://zuul.opendev.org/t/openstack/build/6bcae46ce3594296b7720f29accb6ada : SUCCESS in 28m 54s\n- openstack-ansible-deploy-hosts_lxc_overlayfs-debian-bookworm https://zuul.opendev.org/t/openstack/build/c8c2a070c78643178221d80e70206cf7 : SUCCESS in 31m 56s\n- openstack-ansible-deploy-hosts_lxc_overlayfs-rockylinux-10 https://zuul.opendev.org/t/openstack/build/19c944a965ef4efbaafed19b4ea47367 : SUCCESS in 34m 08s\n- openstack-ansible-deploy-hosts_lxc_btrfs-debian-trixie https://zuul.opendev.org/t/openstack/build/a039cc516f7e4bb08f5a797f5b88d2aa : SUCCESS in 29m 56s\n- openstack-ansible-deploy-hosts_lxc_overlayfs-debian-trixie https://zuul.opendev.org/t/openstack/build/88ccc7b1146e4f23b915d41176c39dd0 : SUCCESS in 27m 56s","accounts_in_message":[],"_revision_number":1},{"id":"41dd0ed526adff1f07ec65b8402c106fa7aa656e","author":{"_account_id":31542,"name":"Andrew Bonney","email":"andrew.bonney@bbc.co.uk","username":"andrewbonney"},"date":"2026-05-27 08:37:43.000000000","message":"Patch Set 1: Code-Review+2","accounts_in_message":[],"_revision_number":1},{"id":"88b499467cce3ba701ad39220f4311dad84af6e8","author":{"_account_id":25023,"name":"Jonathan Rosser","email":"jonathan.rosser@rd.bbc.co.uk","username":"jrosser"},"date":"2026-05-28 17:18:42.000000000","message":"Patch Set 1: Code-Review+2 Workflow+1","accounts_in_message":[],"_revision_number":1},{"id":"56b77214bad75c38fda17af68031596b3ab74aaf","tag":"autogenerated:zuul:gate","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2026-05-28 17:21:16.000000000","message":"Patch Set 1: -Verified\n\nStarting gate jobs.","accounts_in_message":[],"_revision_number":1},{"id":"550cfd3774e7c2a1cf3cc494b4b39f7517ae5fc5","tag":"autogenerated:zuul:gate","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2026-05-28 19:26:56.000000000","message":"Patch Set 1: Verified-2\n\nBuild failed (gate pipeline).  For information on how to proceed, see\nhttps://docs.opendev.org/opendev/infra-manual/latest/developers.html#automated-testing\n\nhttps://zuul.opendev.org/t/openstack/buildset/51a759471b564fcb8eaf272238e9723a\n\n- openstack-ansible-linters-ubuntu-noble https://zuul.opendev.org/t/openstack/build/3e0f144a2ad946dda4c3bccf5bea78e0 : SUCCESS in 10m 22s\n- openstack-tox-docs https://zuul.opendev.org/t/openstack/build/91102fb514644798a4f59eda238ed54d : SUCCESS in 4m 24s\n- build-openstack-releasenotes https://zuul.opendev.org/t/openstack/build/199487a1b857496fa6ba1d3e9dd92748 : SUCCESS in 6m 09s\n- openstack-ansible-deploy-infra_lxc-centos-10-stream https://zuul.opendev.org/t/openstack/build/23c193feac5c4cd9a84aa58517c9a051 : FAILURE in 2h 01m 57s\n- openstack-ansible-deploy-infra_lxc-debian-bookworm https://zuul.opendev.org/t/openstack/build/074a7b3b64124391acec148b0b4e9faa : SUCCESS in 1h 41m 02s\n- openstack-ansible-deploy-infra_lxc-debian-trixie https://zuul.opendev.org/t/openstack/build/a69a32c3ac8c46a19853a8d1296554f2 : SUCCESS in 1h 38m 13s\n- openstack-ansible-deploy-infra_lxc-rockylinux-10 https://zuul.opendev.org/t/openstack/build/ab17c31456844b6799c6d088f32c0451 : SUCCESS in 1h 10m 26s\n- openstack-ansible-deploy-infra_lxc_tls-rockylinux-10 https://zuul.opendev.org/t/openstack/build/870f684164d045389a53a6ca5cac9139 : SUCCESS in 1h 12m 04s\n- openstack-ansible-deploy-infra_lxc-ubuntu-noble https://zuul.opendev.org/t/openstack/build/52cbd28cbfc84ed5ab914c3aeed75de5 : SUCCESS in 1h 02m 36s\n- openstack-ansible-deploy-infra_lxc_stepca-ubuntu-noble https://zuul.opendev.org/t/openstack/build/47ae829c8b17476583cb9950250179dd : SUCCESS in 2h 00m 41s\n- openstack-ansible-deploy-infra_lxc_tls-ubuntu-noble https://zuul.opendev.org/t/openstack/build/81383fb3582d41feb1090ef0a882ce5f : SUCCESS in 1h 46m 42s\n- openstack-ansible-upgrade-infra_lxc-ubuntu-noble https://zuul.opendev.org/t/openstack/build/2dd25277de224d44a288a46ab6234e2b : FAILURE in 1h 14m 35s\n- openstack-ansible-deploy-hosts_lxc_zfs-ubuntu-noble https://zuul.opendev.org/t/openstack/build/c72532e347bc43a195ae2ecc3a0ea2fb : SUCCESS in 16m 58s\n- openstack-ansible-deploy-hosts_lxc_btrfs-ubuntu-noble https://zuul.opendev.org/t/openstack/build/a9a90d5040cb46dea49d6612fb7ce89e : SUCCESS in 27m 34s\n- openstack-ansible-deploy-hosts_lxc_overlayfs-ubuntu-noble https://zuul.opendev.org/t/openstack/build/2b02f4d428794dc4927bf2e638860370 : SUCCESS in 18m 15s\n- openstack-ansible-deploy-hosts_lxc_btrfs-debian-bookworm https://zuul.opendev.org/t/openstack/build/e810ea84bb6f427ab786e1e6db2bce30 : SUCCESS in 25m 08s\n- openstack-ansible-deploy-hosts_lxc_overlayfs-debian-bookworm https://zuul.opendev.org/t/openstack/build/5aaaaf6562954074b7db41902c898dec : SUCCESS in 24m 02s\n- openstack-ansible-deploy-hosts_lxc_btrfs-debian-trixie https://zuul.opendev.org/t/openstack/build/e5f95d841a9b4798aad6990264fbf501 : SUCCESS in 21m 32s\n- openstack-ansible-deploy-hosts_lxc_overlayfs-debian-trixie https://zuul.opendev.org/t/openstack/build/5dbb7b145bae4010bf86432f1b200bfc : SUCCESS in 24m 33s\n- openstack-ansible-deploy-hosts_lxc_overlayfs-rockylinux-10 https://zuul.opendev.org/t/openstack/build/496d21535a924c278e00b96d0263cce8 : SUCCESS in 31m 59s","accounts_in_message":[],"_revision_number":1},{"id":"5b439ad641aebb196fdb86a96dd5729d30c581e4","author":{"_account_id":28619,"name":"Dmitriy Rabotyagov","email":"noonedeadpunk@gmail.com","username":"noonedeadpunk"},"date":"2026-05-29 09:55:08.000000000","message":"Patch Set 1:\n\n(1 comment)","accounts_in_message":[],"_revision_number":1},{"id":"558b9375753663a6d638a56d782cd765cb2e3615","tag":"autogenerated:zuul:check","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2026-05-29 12:19:01.000000000","message":"Patch Set 1: Verified-1\n\nBuild failed (check pipeline).  For information on how to proceed, see\nhttps://docs.opendev.org/opendev/infra-manual/latest/developers.html#automated-testing\nand https://docs.openstack.org/project-team-guide/testing.html#how-to-handle-test-failures\n\nhttps://zuul.opendev.org/t/openstack/buildset/74929ca56a0c402db01af56543cfff3a\n\n- openstack-ansible-linters-ubuntu-noble https://zuul.opendev.org/t/openstack/build/f5891d26e5f24761b0bdd3c2e2edb042 : SUCCESS in 11m 26s\n- openstack-tox-docs https://zuul.opendev.org/t/openstack/build/63d42dad90b14aae98e3550b067f3cf3 : SUCCESS in 5m 52s\n- build-openstack-releasenotes https://zuul.opendev.org/t/openstack/build/07c5c8ebaf474535b48f5ff169e2759f : SUCCESS in 4m 01s\n- openstack-ansible-deploy-infra_lxc-centos-10-stream https://zuul.opendev.org/t/openstack/build/cb7cad841b2b4e73b89ea15252dd82be : FAILURE in 1h 54m 05s\n- openstack-ansible-deploy-infra_lxc-debian-bookworm https://zuul.opendev.org/t/openstack/build/b98203890a294cbca1efe5a55d827df3 : SUCCESS in 2h 00m 03s\n- openstack-ansible-deploy-infra_lxc-debian-trixie https://zuul.opendev.org/t/openstack/build/7a90008d3d3a4fa59410a88e80081b55 : SUCCESS in 1h 55m 07s\n- openstack-ansible-deploy-infra_lxc-rockylinux-10 https://zuul.opendev.org/t/openstack/build/4ba12dd7481443b4b104e22ae43a27e5 : SUCCESS in 1h 18m 11s\n- openstack-ansible-deploy-infra_lxc_tls-rockylinux-10 https://zuul.opendev.org/t/openstack/build/a26cdbfd341a4cb58a9637090648b6ab : SUCCESS in 1h 49m 58s\n- openstack-ansible-deploy-infra_lxc-ubuntu-noble https://zuul.opendev.org/t/openstack/build/f6d0a733f7f04889b81f8cfc3ce660db : FAILURE in 1h 33m 33s\n- openstack-ansible-deploy-infra_lxc_stepca-ubuntu-noble https://zuul.opendev.org/t/openstack/build/ef9b9f04c1ae4ca79cda9223a8da5534 : SUCCESS in 1h 49m 38s\n- openstack-ansible-deploy-infra_lxc_tls-ubuntu-noble https://zuul.opendev.org/t/openstack/build/6f539a43b285487b984961c8c1189ce3 : SUCCESS in 2h 14m 46s\n- openstack-ansible-upgrade-infra_lxc-ubuntu-noble https://zuul.opendev.org/t/openstack/build/0a14fee58b9d441aabf001d6e83c2afd : SUCCESS in 2h 16m 34s\n- openstack-ansible-deploy-hosts_lxc_zfs-ubuntu-noble https://zuul.opendev.org/t/openstack/build/8a12578b151e49489f7c21094bcd3619 : SUCCESS in 29m 29s\n- openstack-ansible-deploy-hosts_lxc_btrfs-ubuntu-noble https://zuul.opendev.org/t/openstack/build/b7591d8cf4de4733b9261269d9953b66 : SUCCESS in 27m 42s\n- openstack-ansible-deploy-hosts_lxc_overlayfs-ubuntu-noble https://zuul.opendev.org/t/openstack/build/ee4e3fe4e7c34aba99c7e18f2a96426c : SUCCESS in 16m 47s\n- openstack-ansible-deploy-hosts_lxc_btrfs-debian-bookworm https://zuul.opendev.org/t/openstack/build/1b94ccb074b344b884377a0cb34670d7 : SUCCESS in 27m 24s\n- openstack-ansible-deploy-hosts_lxc_overlayfs-debian-bookworm https://zuul.opendev.org/t/openstack/build/1c43296dcda0463aaeb5322851e95cb5 : SUCCESS in 31m 35s\n- openstack-ansible-deploy-hosts_lxc_overlayfs-rockylinux-10 https://zuul.opendev.org/t/openstack/build/d0078c234b264c4ab1f493807c49f660 : SUCCESS in 25m 59s\n- openstack-ansible-deploy-hosts_lxc_btrfs-debian-trixie https://zuul.opendev.org/t/openstack/build/1d6f4f4d71864e678ecfd62842d5aa7b : SUCCESS in 24m 43s\n- openstack-ansible-deploy-hosts_lxc_overlayfs-debian-trixie https://zuul.opendev.org/t/openstack/build/16ea271bfd4343f99a40049c9ad2c498 : SUCCESS in 27m 31s","accounts_in_message":[],"_revision_number":1},{"id":"cc98bb7effc1e3a234880286381bb6b7f05ff262","author":{"_account_id":37598,"name":"Ivan Anfimov","display_name":"Ivan Anfimov","email":"lazekteam@gmail.com","username":"anfimovir"},"date":"2026-05-30 20:26:08.000000000","message":"Patch Set 1:\n\n(1 comment)","accounts_in_message":[],"_revision_number":1},{"id":"3ccf33efe853a8c189e5157bc7a197074d99949d","tag":"autogenerated:zuul:check","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2026-05-30 22:45:05.000000000","message":"Patch Set 1:\n\nBuild failed (check pipeline).  For information on how to proceed, see\nhttps://docs.opendev.org/opendev/infra-manual/latest/developers.html#automated-testing\nand https://docs.openstack.org/project-team-guide/testing.html#how-to-handle-test-failures\n\nhttps://zuul.opendev.org/t/openstack/buildset/3f4f50264fde405a92588d24c2b08193\n\n- openstack-ansible-linters-ubuntu-noble https://zuul.opendev.org/t/openstack/build/5558bf1d5902423d95629459b7ea08ab : SUCCESS in 11m 21s\n- openstack-tox-docs https://zuul.opendev.org/t/openstack/build/1d2111f4505d424093639ee9e89be001 : SUCCESS in 4m 36s\n- build-openstack-releasenotes https://zuul.opendev.org/t/openstack/build/af8e75034dde4e988ccac56469fe6951 : SUCCESS in 2m 57s\n- openstack-ansible-deploy-infra_lxc-centos-10-stream https://zuul.opendev.org/t/openstack/build/c934d15551e74e4c87b7181ff9c18b8f : FAILURE in 1h 53m 36s\n- openstack-ansible-deploy-infra_lxc-debian-bookworm https://zuul.opendev.org/t/openstack/build/f6aa11b167344673b41d13538dd8d26d : SUCCESS in 1h 51m 20s\n- openstack-ansible-deploy-infra_lxc-debian-trixie https://zuul.opendev.org/t/openstack/build/34a2961be0ce4c15941ceafac0a353fc : SUCCESS in 1h 53m 02s\n- openstack-ansible-deploy-infra_lxc-rockylinux-10 https://zuul.opendev.org/t/openstack/build/0b1736c0fec641e0858bb3501863d414 : SUCCESS in 1h 39m 20s\n- openstack-ansible-deploy-infra_lxc_tls-rockylinux-10 https://zuul.opendev.org/t/openstack/build/d75ef0ca5b234840b23d27dc6ddb3309 : SUCCESS in 1h 39m 06s\n- openstack-ansible-deploy-infra_lxc-ubuntu-noble https://zuul.opendev.org/t/openstack/build/9fdcb8f19e4842aa9bf666682330098e : SUCCESS in 1h 54m 31s\n- openstack-ansible-deploy-infra_lxc_stepca-ubuntu-noble https://zuul.opendev.org/t/openstack/build/aa1a19d5cc0942a2887bf628e6758d75 : SUCCESS in 59m 30s\n- openstack-ansible-deploy-infra_lxc_tls-ubuntu-noble https://zuul.opendev.org/t/openstack/build/2e47ba575fee4687928398b65306fa3f : SUCCESS in 1h 32m 46s\n- openstack-ansible-upgrade-infra_lxc-ubuntu-noble https://zuul.opendev.org/t/openstack/build/251a5eb6133943e58b23aa905ce48959 : SUCCESS in 2h 15m 13s\n- openstack-ansible-deploy-hosts_lxc_zfs-ubuntu-noble https://zuul.opendev.org/t/openstack/build/097b0d1b570347dc8c8f9b93e61bc321 : SUCCESS in 20m 05s\n- openstack-ansible-deploy-hosts_lxc_btrfs-ubuntu-noble https://zuul.opendev.org/t/openstack/build/6d8362bb98fe4d3e8f110380fdf3f0b9 : SUCCESS in 18m 04s\n- openstack-ansible-deploy-hosts_lxc_overlayfs-ubuntu-noble https://zuul.opendev.org/t/openstack/build/59760fe37ecf4bb3b3a52ed0be276c07 : SUCCESS in 26m 33s\n- openstack-ansible-deploy-hosts_lxc_btrfs-debian-bookworm https://zuul.opendev.org/t/openstack/build/4f5cc00c7f8d46e98906dd28ffc90e70 : SUCCESS in 24m 00s\n- openstack-ansible-deploy-hosts_lxc_overlayfs-debian-bookworm https://zuul.opendev.org/t/openstack/build/cae03127437a4424bf35c4eb33a0e9c1 : SUCCESS in 24m 07s\n- openstack-ansible-deploy-hosts_lxc_overlayfs-rockylinux-10 https://zuul.opendev.org/t/openstack/build/06ba48c4ce6b49a69189fd880e059db3 : SUCCESS in 22m 09s\n- openstack-ansible-deploy-hosts_lxc_btrfs-debian-trixie https://zuul.opendev.org/t/openstack/build/1230864dc8974df89a6c65c542b66a72 : SUCCESS in 30m 51s\n- openstack-ansible-deploy-hosts_lxc_overlayfs-debian-trixie https://zuul.opendev.org/t/openstack/build/3e7144a7a05a46c6a3227cbf40e7f2e4 : SUCCESS in 30m 31s","accounts_in_message":[],"_revision_number":1},{"id":"d5127fa7fdef5af41962fa9bad9fa12b32b32c79","author":{"_account_id":28619,"name":"Dmitriy Rabotyagov","email":"noonedeadpunk@gmail.com","username":"noonedeadpunk"},"date":"2026-06-01 11:54:50.000000000","message":"Patch Set 1:\n\n(1 comment)","accounts_in_message":[],"_revision_number":1},{"id":"aacefe73443ee2c1250b82e6ba4ff8ade20b7cbc","tag":"autogenerated:zuul:check","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2026-06-01 13:31:31.000000000","message":"Patch Set 1:\n\nBuild failed (check pipeline).  For information on how to proceed, see\nhttps://docs.opendev.org/opendev/infra-manual/latest/developers.html#automated-testing\nand https://docs.openstack.org/project-team-guide/testing.html#how-to-handle-test-failures\n\nhttps://zuul.opendev.org/t/openstack/buildset/53e532e6992c4ca2a1348fa6edce5a81\n\n- openstack-ansible-linters-ubuntu-noble https://zuul.opendev.org/t/openstack/build/4769cd6348494512b36ec4d714b3990b : SUCCESS in 13m 01s\n- openstack-tox-docs https://zuul.opendev.org/t/openstack/build/99bf4535921741028c291f984900aa08 : SUCCESS in 6m 17s\n- build-openstack-releasenotes https://zuul.opendev.org/t/openstack/build/bfccaa46c7d94a90b78c57cc246d7d70 : SUCCESS in 3m 24s\n- openstack-ansible-deploy-infra_lxc-centos-10-stream https://zuul.opendev.org/t/openstack/build/aa3306e9734f44f3b180db1a6adf228b : FAILURE in 53m 25s\n- openstack-ansible-deploy-infra_lxc-debian-bookworm https://zuul.opendev.org/t/openstack/build/b5c51bf94e7e42719d41f3510c0de991 : FAILURE in 1h 09m 32s\n- openstack-ansible-deploy-infra_lxc-debian-trixie https://zuul.opendev.org/t/openstack/build/e87f1f0c6b8748fab3918d6e8ae6e084 : FAILURE in 54m 26s\n- openstack-ansible-deploy-infra_lxc-rockylinux-10 https://zuul.opendev.org/t/openstack/build/a1a922150ded40b7ab9e08031e51d2ce : FAILURE in 58m 58s\n- openstack-ansible-deploy-infra_lxc_tls-rockylinux-10 https://zuul.opendev.org/t/openstack/build/91c551b7c98e4840b77418f048452d9f : FAILURE in 1h 00m 44s\n- openstack-ansible-deploy-infra_lxc-ubuntu-noble https://zuul.opendev.org/t/openstack/build/eb1f13cde79947fcb63dd11963e4bd8f : FAILURE in 1h 13m 37s\n- openstack-ansible-deploy-infra_lxc_stepca-ubuntu-noble https://zuul.opendev.org/t/openstack/build/d05db8de777a4d6e8045bc11af8dac17 : FAILURE in 1h 23m 05s\n- openstack-ansible-deploy-infra_lxc_tls-ubuntu-noble https://zuul.opendev.org/t/openstack/build/ac0dc2f590514095953045be57d12975 : FAILURE in 1h 09m 06s\n- openstack-ansible-upgrade-infra_lxc-ubuntu-noble https://zuul.opendev.org/t/openstack/build/aebf29eb7d634395b4bb48647b08a770 : FAILURE in 1h 19m 30s\n- openstack-ansible-deploy-hosts_lxc_zfs-ubuntu-noble https://zuul.opendev.org/t/openstack/build/5480202398484c299d55d114be9cfde6 : FAILURE in 28m 53s\n- openstack-ansible-deploy-hosts_lxc_btrfs-ubuntu-noble https://zuul.opendev.org/t/openstack/build/cd4e73153aff4920854798849d20c479 : FAILURE in 29m 09s\n- openstack-ansible-deploy-hosts_lxc_overlayfs-ubuntu-noble https://zuul.opendev.org/t/openstack/build/31d8969372f6479b811ac130ccb299c8 : FAILURE in 34m 42s\n- openstack-ansible-deploy-hosts_lxc_btrfs-debian-bookworm https://zuul.opendev.org/t/openstack/build/9429e049ddfa44cabdb5c2659fdfee7e : FAILURE in 27m 02s\n- openstack-ansible-deploy-hosts_lxc_overlayfs-debian-bookworm https://zuul.opendev.org/t/openstack/build/795c0d3d38a141c7ae792d37c1cf6e25 : FAILURE in 24m 16s\n- openstack-ansible-deploy-hosts_lxc_overlayfs-rockylinux-10 https://zuul.opendev.org/t/openstack/build/5e91b3f5a56d4003af0be2cd0e0875d0 : FAILURE in 29m 22s\n- openstack-ansible-deploy-hosts_lxc_btrfs-debian-trixie https://zuul.opendev.org/t/openstack/build/ffbd329783a844038de578aea8c0d1b4 : FAILURE in 26m 26s\n- openstack-ansible-deploy-hosts_lxc_overlayfs-debian-trixie https://zuul.opendev.org/t/openstack/build/f8ac951c67b946dabec6dcc51c56d62e : FAILURE in 28m 25s","accounts_in_message":[],"_revision_number":1},{"id":"80c63d124931b815be71ed9808d5d82219a2fec8","author":{"_account_id":28619,"name":"Dmitriy Rabotyagov","email":"noonedeadpunk@gmail.com","username":"noonedeadpunk"},"date":"2026-06-01 21:38:11.000000000","message":"Patch Set 1:\n\n(1 comment)","accounts_in_message":[],"_revision_number":1},{"id":"84f9fe8b628793d8443f27c5695b179a59e389a8","tag":"autogenerated:zuul:check","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2026-06-01 23:29:54.000000000","message":"Patch Set 1:\n\nBuild failed (check pipeline).  For information on how to proceed, see\nhttps://docs.opendev.org/opendev/infra-manual/latest/developers.html#automated-testing\nand https://docs.openstack.org/project-team-guide/testing.html#how-to-handle-test-failures\n\nhttps://zuul.opendev.org/t/openstack/buildset/1d2947cfcc664523817cb1938071f2b6\n\n- openstack-ansible-linters-ubuntu-noble https://zuul.opendev.org/t/openstack/build/863833abaa79496b82822ee1affa9f4b : SUCCESS in 12m 28s\n- openstack-tox-docs https://zuul.opendev.org/t/openstack/build/7e40e37b61224cd581374a17efe9eab9 : SUCCESS in 6m 13s\n- build-openstack-releasenotes https://zuul.opendev.org/t/openstack/build/745c74fdc8c04d828f22789395b36938 : SUCCESS in 3m 43s\n- openstack-ansible-deploy-infra_lxc-centos-10-stream https://zuul.opendev.org/t/openstack/build/d42ac11facb44141a89f46a9c29301ce : SUCCESS in 1h 01m 42s\n- openstack-ansible-deploy-infra_lxc-debian-bookworm https://zuul.opendev.org/t/openstack/build/2b1b73777d134c5fa91bea95257f8b11 : SUCCESS in 1h 33m 21s\n- openstack-ansible-deploy-infra_lxc-debian-trixie https://zuul.opendev.org/t/openstack/build/aa19cea9b49344f383ff368f9f7b03f1 : SUCCESS in 56m 10s\n- openstack-ansible-deploy-infra_lxc-rockylinux-10 https://zuul.opendev.org/t/openstack/build/4e440ac8b2b142e3b60d2d2c1756c8f6 : SUCCESS in 1h 01m 10s\n- openstack-ansible-deploy-infra_lxc_tls-rockylinux-10 https://zuul.opendev.org/t/openstack/build/2c746aae34404b14a70700db29ddcfc8 : SUCCESS in 1h 00m 36s\n- openstack-ansible-deploy-infra_lxc-ubuntu-noble https://zuul.opendev.org/t/openstack/build/fc671497bd7d41f181d4474d1206fbc1 : SUCCESS in 1h 31m 24s\n- openstack-ansible-deploy-infra_lxc_stepca-ubuntu-noble https://zuul.opendev.org/t/openstack/build/e12e389b2a81480e8a8c0dde4331415e : SUCCESS in 1h 40m 14s\n- openstack-ansible-deploy-infra_lxc_tls-ubuntu-noble https://zuul.opendev.org/t/openstack/build/1afadc83be444df78c0b376f8163f79c : SUCCESS in 1h 37m 12s\n- openstack-ansible-upgrade-infra_lxc-ubuntu-noble https://zuul.opendev.org/t/openstack/build/0f825f15301f4a08a3248cb95675195d : FAILURE in 14m 37s\n- openstack-ansible-deploy-hosts_lxc_zfs-ubuntu-noble https://zuul.opendev.org/t/openstack/build/0a02bcb00f504d66b44f95a9197a0b58 : FAILURE in 29m 12s\n- openstack-ansible-deploy-hosts_lxc_btrfs-ubuntu-noble https://zuul.opendev.org/t/openstack/build/1fd72a125470402da8373252a402f92e : FAILURE in 27m 38s\n- openstack-ansible-deploy-hosts_lxc_overlayfs-ubuntu-noble https://zuul.opendev.org/t/openstack/build/0504e9cd3af24a59a3d60531f0aa493b : FAILURE in 26m 40s\n- openstack-ansible-deploy-hosts_lxc_btrfs-debian-bookworm https://zuul.opendev.org/t/openstack/build/098e295fe24d4788957e476764106ae0 : FAILURE in 27m 28s\n- openstack-ansible-deploy-hosts_lxc_overlayfs-debian-bookworm https://zuul.opendev.org/t/openstack/build/f941681205a14493accb923d287fe09f : FAILURE in 28m 04s\n- openstack-ansible-deploy-hosts_lxc_overlayfs-rockylinux-10 https://zuul.opendev.org/t/openstack/build/299e524bc90b45029cb18bf588391ebf : FAILURE in 21m 05s\n- openstack-ansible-deploy-hosts_lxc_btrfs-debian-trixie https://zuul.opendev.org/t/openstack/build/a657aacdce7f4843a8ef2eeb04b4f5cd : FAILURE in 31m 12s\n- openstack-ansible-deploy-hosts_lxc_overlayfs-debian-trixie https://zuul.opendev.org/t/openstack/build/7521c2f7766845d1aafcfda0f05c81ca : FAILURE in 27m 49s","accounts_in_message":[],"_revision_number":1},{"id":"d40496ea31d3c84c9ec3e095cde952329422318f","tag":"autogenerated:gerrit:newPatchSet","author":{"_account_id":28619,"name":"Dmitriy Rabotyagov","email":"noonedeadpunk@gmail.com","username":"noonedeadpunk"},"date":"2026-06-02 05:52:00.000000000","message":"Patch Set 2: Commit message was updated.\n\nOutdated Votes:\n* Code-Review+2 (copy condition: \"changekind:TRIVIAL_REBASE OR is:MIN\")\n* Verified-1 (copy condition: \"NEVER\")\n* Workflow+1 (copy condition: \"NEVER\")\n","accounts_in_message":[],"_revision_number":2},{"id":"ec7611446c46a27c6473795969639003d6684df3","tag":"autogenerated:zuul:check","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2026-06-02 06:07:31.000000000","message":"Patch Set 2: Verified-1\n\nThis change depends on a change that failed to merge.\n\nChange https://review.opendev.org/c/openstack/openstack-ansible-plugins/+/991016 is needed.","accounts_in_message":[],"_revision_number":2},{"id":"f794b81e872c2e5601598c09ca9e5d8bbd854ed2","author":{"_account_id":28619,"name":"Dmitriy Rabotyagov","email":"noonedeadpunk@gmail.com","username":"noonedeadpunk"},"date":"2026-06-02 06:07:50.000000000","message":"Patch Set 2:\n\n(1 comment)","accounts_in_message":[],"_revision_number":2},{"id":"b4792d5b1c85d4995a05288ad81fc7706e042183","tag":"autogenerated:zuul:check","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2026-06-02 08:30:56.000000000","message":"Patch Set 2: Verified+1\n\nBuild succeeded (check pipeline).\nhttps://zuul.opendev.org/t/openstack/buildset/83960ff01e32418182a42049974c4b54\n\n- openstack-ansible-linters-ubuntu-noble https://zuul.opendev.org/t/openstack/build/de85f6331fd0442b9eb77a3cb5462823 : SUCCESS in 12m 25s\n- openstack-tox-docs https://zuul.opendev.org/t/openstack/build/61579fe180b840fb9ec8f55a93e9df5c : SUCCESS in 5m 27s\n- build-openstack-releasenotes https://zuul.opendev.org/t/openstack/build/2911b9efceb840d78ee1957d432a3488 : SUCCESS in 3m 31s\n- openstack-ansible-deploy-infra_lxc-centos-10-stream https://zuul.opendev.org/t/openstack/build/36b665215e16441fa822ff8e6aa4474e : SUCCESS in 1h 13m 16s\n- openstack-ansible-deploy-infra_lxc-debian-bookworm https://zuul.opendev.org/t/openstack/build/86c13f5bc15d4d85844e96935088aafc : SUCCESS in 1h 36m 04s\n- openstack-ansible-deploy-infra_lxc-debian-trixie https://zuul.opendev.org/t/openstack/build/39e053e18fec428aad93e4556a82f94d : SUCCESS in 1h 53m 11s\n- openstack-ansible-deploy-infra_lxc-rockylinux-10 https://zuul.opendev.org/t/openstack/build/e8c7106045f94582a13f65463154f9aa : SUCCESS in 1h 43m 15s\n- openstack-ansible-deploy-infra_lxc_tls-rockylinux-10 https://zuul.opendev.org/t/openstack/build/17a0bd8220044f74beea645dd9540061 : SUCCESS in 1h 46m 57s\n- openstack-ansible-deploy-infra_lxc-ubuntu-noble https://zuul.opendev.org/t/openstack/build/186534b62ef7457990023875d4220bc7 : SUCCESS in 1h 35m 16s\n- openstack-ansible-deploy-infra_lxc_stepca-ubuntu-noble https://zuul.opendev.org/t/openstack/build/208bd081f8f2480eaa7cebb8c5168dcb : SUCCESS in 1h 37m 21s\n- openstack-ansible-deploy-infra_lxc_tls-ubuntu-noble https://zuul.opendev.org/t/openstack/build/72a95125130941268b2172e84305c090 : SUCCESS in 1h 35m 50s\n- openstack-ansible-upgrade-infra_lxc-ubuntu-noble https://zuul.opendev.org/t/openstack/build/aae11cbfaf64435884a70147635c931c : SUCCESS in 2h 04m 12s\n- openstack-ansible-deploy-hosts_lxc_zfs-ubuntu-noble https://zuul.opendev.org/t/openstack/build/68ce5c088f7e439ea84fa87e725b9ecb : SUCCESS in 26m 19s\n- openstack-ansible-deploy-hosts_lxc_btrfs-ubuntu-noble https://zuul.opendev.org/t/openstack/build/b1136b35073249acbc2882ca90eca3a2 : SUCCESS in 26m 55s\n- openstack-ansible-deploy-hosts_lxc_overlayfs-ubuntu-noble https://zuul.opendev.org/t/openstack/build/f3310438fe5c484f997cd16e5cec7790 : SUCCESS in 26m 11s\n- openstack-ansible-deploy-hosts_lxc_btrfs-debian-bookworm https://zuul.opendev.org/t/openstack/build/511d48d89bd4463f9326108c30993442 : SUCCESS in 27m 44s\n- openstack-ansible-deploy-hosts_lxc_overlayfs-debian-bookworm https://zuul.opendev.org/t/openstack/build/4a603aefdf884b2da44c3dd1adefd59e : SUCCESS in 29m 13s\n- openstack-ansible-deploy-hosts_lxc_overlayfs-rockylinux-10 https://zuul.opendev.org/t/openstack/build/89629fb583804fb89342a2ab88404b3d : SUCCESS in 32m 52s\n- openstack-ansible-deploy-hosts_lxc_btrfs-debian-trixie https://zuul.opendev.org/t/openstack/build/214295ad0e7041fa9d0c398b7a5b1590 : SUCCESS in 27m 20s\n- openstack-ansible-deploy-hosts_lxc_overlayfs-debian-trixie https://zuul.opendev.org/t/openstack/build/94c32ff6f67040b0ad42ea49b1cb3bb2 : SUCCESS in 26m 18s","accounts_in_message":[],"_revision_number":2},{"id":"5959ef0ad8da620f856bf8b0cdfa6ccf24a270c5","author":{"_account_id":31542,"name":"Andrew Bonney","email":"andrew.bonney@bbc.co.uk","username":"andrewbonney"},"date":"2026-06-02 12:46:57.000000000","message":"Patch Set 2: Code-Review+2","accounts_in_message":[],"_revision_number":2},{"id":"0d6dd68fc292b90b7811d7923c61ddedb9496136","author":{"_account_id":37598,"name":"Ivan Anfimov","display_name":"Ivan Anfimov","email":"lazekteam@gmail.com","username":"anfimovir"},"date":"2026-06-02 12:51:53.000000000","message":"Patch Set 2: Code-Review+2","accounts_in_message":[],"_revision_number":2},{"id":"11a2a45d8a7ad59b364691ddf9fc5c6e8d1c8a78","author":{"_account_id":28619,"name":"Dmitriy Rabotyagov","email":"noonedeadpunk@gmail.com","username":"noonedeadpunk"},"date":"2026-06-02 15:03:11.000000000","message":"Patch Set 2: Workflow+1","accounts_in_message":[],"_revision_number":2},{"id":"952cda7975c2d50a99835efff7a47083938150ee","tag":"autogenerated:zuul:gate","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2026-06-02 15:06:06.000000000","message":"Patch Set 2: -Verified\n\nStarting gate jobs.","accounts_in_message":[],"_revision_number":2},{"id":"6cefa6e655b6ee00fcd5f39ca428895337e1c107","tag":"autogenerated:zuul:gate","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2026-06-02 17:31:34.000000000","message":"Patch Set 2: Verified+2\n\nBuild succeeded (gate pipeline).\nhttps://zuul.opendev.org/t/openstack/buildset/e8e5b24a4dea4a37afadc0ce6df406a2\n\n- openstack-ansible-linters-ubuntu-noble https://zuul.opendev.org/t/openstack/build/a85b0b54aaba4f7eb5fc1a852ce9f45a : SUCCESS in 18m 43s\n- openstack-tox-docs https://zuul.opendev.org/t/openstack/build/1a1b6c362286474a88acf89dba6ceffc : SUCCESS in 7m 24s\n- build-openstack-releasenotes https://zuul.opendev.org/t/openstack/build/ed90ef42e56147719185d559c0b7689b : SUCCESS in 5m 56s\n- openstack-ansible-deploy-infra_lxc-centos-10-stream https://zuul.opendev.org/t/openstack/build/87621c48fc364528a17c683ca0fd67c9 : SUCCESS in 1h 36m 38s\n- openstack-ansible-deploy-infra_lxc-debian-bookworm https://zuul.opendev.org/t/openstack/build/ed9a167a3e0e4d989a3bbe241974e9b3 : SUCCESS in 1h 00m 37s\n- openstack-ansible-deploy-infra_lxc-debian-trixie https://zuul.opendev.org/t/openstack/build/3c5a88c2056c4c119cfc337154f6694c : SUCCESS in 1h 48m 48s\n- openstack-ansible-deploy-infra_lxc-rockylinux-10 https://zuul.opendev.org/t/openstack/build/784a7a5003b746009f58ea0162cc306b : SUCCESS in 1h 08m 39s\n- openstack-ansible-deploy-infra_lxc_tls-rockylinux-10 https://zuul.opendev.org/t/openstack/build/1e1e636dd4f9447d952e3dd47080c9c6 : SUCCESS in 1h 53m 31s\n- openstack-ansible-deploy-infra_lxc-ubuntu-noble https://zuul.opendev.org/t/openstack/build/37454bb2044b4075a382762ca83c0013 : SUCCESS in 1h 43m 54s\n- openstack-ansible-deploy-infra_lxc_stepca-ubuntu-noble https://zuul.opendev.org/t/openstack/build/4975cc0f992f4d5f9654fdc06fb54d26 : SUCCESS in 1h 35m 39s\n- openstack-ansible-deploy-infra_lxc_tls-ubuntu-noble https://zuul.opendev.org/t/openstack/build/a12931eb146546108b2d2149c7336d69 : SUCCESS in 58m 33s\n- openstack-ansible-upgrade-infra_lxc-ubuntu-noble https://zuul.opendev.org/t/openstack/build/8ebbf1f3593644c6b908111452e8eb68 : SUCCESS in 2h 19m 59s\n- openstack-ansible-deploy-hosts_lxc_zfs-ubuntu-noble https://zuul.opendev.org/t/openstack/build/7a5ea43cb3d34f4393ca0709cf8d41be : SUCCESS in 28m 50s\n- openstack-ansible-deploy-hosts_lxc_btrfs-ubuntu-noble https://zuul.opendev.org/t/openstack/build/7840ffdbd90946a7b7d0330846a3173d : SUCCESS in 23m 20s\n- openstack-ansible-deploy-hosts_lxc_overlayfs-ubuntu-noble https://zuul.opendev.org/t/openstack/build/ffaad97090d0404090d2d358f2ee1038 : SUCCESS in 27m 07s\n- openstack-ansible-deploy-hosts_lxc_btrfs-debian-bookworm https://zuul.opendev.org/t/openstack/build/0589223fa80d42ab8752a246d3785e45 : SUCCESS in 23m 43s\n- openstack-ansible-deploy-hosts_lxc_overlayfs-debian-bookworm https://zuul.opendev.org/t/openstack/build/3f78856fe2ac4b51ace3989341d4ee8b : SUCCESS in 21m 47s\n- openstack-ansible-deploy-hosts_lxc_btrfs-debian-trixie https://zuul.opendev.org/t/openstack/build/0cfedb7e254948faa2ebbbce7223a8fb : SUCCESS in 33m 25s\n- openstack-ansible-deploy-hosts_lxc_overlayfs-debian-trixie https://zuul.opendev.org/t/openstack/build/dd16dbe9e5c046ebb50594ba1a53de67 : SUCCESS in 32m 14s\n- openstack-ansible-deploy-hosts_lxc_overlayfs-rockylinux-10 https://zuul.opendev.org/t/openstack/build/f3090445566d48c296299b579b8ae9c9 : SUCCESS in 22m 46s","accounts_in_message":[],"_revision_number":2},{"id":"b3b6568c435b95246d32a0cc24d22e423d626278","tag":"autogenerated:gerrit:merged","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2026-06-02 17:31:34.000000000","message":"Change has been successfully merged","accounts_in_message":[],"_revision_number":2},{"id":"7ae33803825b5fdc9b9bf7b322cad03d7edc2702","tag":"autogenerated:zuul:promote","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2026-06-02 17:33:13.000000000","message":"Patch Set 2:\n\nBuild succeeded (promote pipeline).\nhttps://zuul.opendev.org/t/openstack/buildset/d1e94a31373c49a39d886edf847f1087\n\n- promote-openstack-tox-docs https://zuul.opendev.org/t/openstack/build/e57fb9cbaadd4be496741c131f68d7b3 : SUCCESS in 1m 13s","accounts_in_message":[],"_revision_number":2}],"current_revision_number":2,"current_revision":"627364f0b0bb3643f95acc055bf162f469b465e6","revisions":{"2cd1fe24b4266670b9053ae0ba40d1fb6a7d323f":{"kind":"REWORK","_number":1,"created":"2026-05-26 09:13:15.000000000","uploader":{"_account_id":28619,"name":"Dmitriy Rabotyagov","email":"noonedeadpunk@gmail.com","username":"noonedeadpunk"},"ref":"refs/changes/70/990070/1","fetch":{"anonymous http":{"url":"https://review.opendev.org/openstack/openstack-ansible-lxc_hosts","ref":"refs/changes/70/990070/1","commands":{"Checkout":"git fetch https://review.opendev.org/openstack/openstack-ansible-lxc_hosts refs/changes/70/990070/1 \u0026\u0026 git checkout FETCH_HEAD","Cherry Pick":"git fetch https://review.opendev.org/openstack/openstack-ansible-lxc_hosts refs/changes/70/990070/1 \u0026\u0026 git cherry-pick FETCH_HEAD","Format Patch":"git fetch https://review.opendev.org/openstack/openstack-ansible-lxc_hosts refs/changes/70/990070/1 \u0026\u0026 git format-patch -1 --stdout FETCH_HEAD","Pull":"git pull https://review.opendev.org/openstack/openstack-ansible-lxc_hosts refs/changes/70/990070/1"}}},"commit":{"parents":[{"commit":"3eab0d7262eb0c79acf305a89ab3e8ebd9c4fe54","subject":"Merge \"Add support for Ubuntu 26.04\"","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/openstack-ansible-lxc_hosts/commit/3eab0d7262eb0c79acf305a89ab3e8ebd9c4fe54"}]}],"author":{"name":"Dmitriy Rabotyagov","email":"dmitriy.rabotyagov@cleura.com","date":"2026-05-26 09:13:05.000000000","tz":120},"committer":{"name":"Dmitriy Rabotyagov","email":"dmitriy.rabotyagov@cleura.com","date":"2026-05-26 09:13:05.000000000","tz":120},"subject":"Ensure ubuntu_virt hook not passed to LXC","message":"Ensure ubuntu_virt hook not passed to LXC\n\nThis removes ubuntu_virt hook which is expecting\n/usr/bin/apt_hook_ubuntu_virt to be present on the system.\n\nAs currently we\u0027re not expecting for QEMU to run inside of the\ncontainers, and even it should - this needs to be handled separately\nrather then include into the base image.\n\nChange-Id: If5f328417ca0c2ec427de70181dd5623c1802029\nSigned-off-by: Dmitriy Rabotyagov \u003cdmitriy.rabotyagov@cleura.com\u003e\n","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/openstack-ansible-lxc_hosts/commit/2cd1fe24b4266670b9053ae0ba40d1fb6a7d323f"}],"resolve_conflicts_web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/openstack-ansible-lxc_hosts/commit/2cd1fe24b4266670b9053ae0ba40d1fb6a7d323f"}]},"branch":"refs/heads/master"},"627364f0b0bb3643f95acc055bf162f469b465e6":{"kind":"NO_CODE_CHANGE","_number":2,"created":"2026-06-02 05:52:00.000000000","uploader":{"_account_id":28619,"name":"Dmitriy Rabotyagov","email":"noonedeadpunk@gmail.com","username":"noonedeadpunk"},"ref":"refs/changes/70/990070/2","fetch":{"anonymous http":{"url":"https://review.opendev.org/openstack/openstack-ansible-lxc_hosts","ref":"refs/changes/70/990070/2","commands":{"Checkout":"git fetch https://review.opendev.org/openstack/openstack-ansible-lxc_hosts refs/changes/70/990070/2 \u0026\u0026 git checkout FETCH_HEAD","Cherry Pick":"git fetch https://review.opendev.org/openstack/openstack-ansible-lxc_hosts refs/changes/70/990070/2 \u0026\u0026 git cherry-pick FETCH_HEAD","Format Patch":"git fetch https://review.opendev.org/openstack/openstack-ansible-lxc_hosts refs/changes/70/990070/2 \u0026\u0026 git format-patch -1 --stdout FETCH_HEAD","Pull":"git pull https://review.opendev.org/openstack/openstack-ansible-lxc_hosts refs/changes/70/990070/2"}}},"commit":{"parents":[{"commit":"3eab0d7262eb0c79acf305a89ab3e8ebd9c4fe54","subject":"Merge \"Add support for Ubuntu 26.04\"","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/openstack-ansible-lxc_hosts/commit/3eab0d7262eb0c79acf305a89ab3e8ebd9c4fe54"}]}],"author":{"name":"Dmitriy Rabotyagov","email":"dmitriy.rabotyagov@cleura.com","date":"2026-05-26 09:13:05.000000000","tz":120},"committer":{"name":"Dmitriy Rabotyagov","email":"dmitriy.rabotyagov@cleura.com","date":"2026-06-02 05:52:00.000000000","tz":0},"subject":"Ensure ubuntu_virt hook not passed to LXC","message":"Ensure ubuntu_virt hook not passed to LXC\n\nThis removes ubuntu_virt hook which is expecting\n/usr/bin/apt_hook_ubuntu_virt to be present on the system.\n\nAs currently we\u0027re not expecting for QEMU to run inside of the\ncontainers, and even it should - this needs to be handled separately\nrather then include into the base image.\n\nDepends-On: https://review.opendev.org/c/openstack/openstack-ansible-plugins/+/991016\nChange-Id: If5f328417ca0c2ec427de70181dd5623c1802029\nSigned-off-by: Dmitriy Rabotyagov \u003cdmitriy.rabotyagov@cleura.com\u003e\n","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/openstack-ansible-lxc_hosts/commit/627364f0b0bb3643f95acc055bf162f469b465e6"}],"resolve_conflicts_web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/openstack-ansible-lxc_hosts/commit/627364f0b0bb3643f95acc055bf162f469b465e6"}]},"branch":"refs/heads/master","description":"Edit commit message"}},"requirements":[],"submit_records":[{"rule_name":"gerrit~DefaultSubmitRule","status":"CLOSED","labels":[{"label":"Verified","status":"MAY","applied_by":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]}},{"label":"Code-Review","status":"MAY","applied_by":{"_account_id":31542,"name":"Andrew Bonney","email":"andrew.bonney@bbc.co.uk","username":"andrewbonney"}},{"label":"Workflow","status":"MAY","applied_by":{"_account_id":28619,"name":"Dmitriy Rabotyagov","email":"noonedeadpunk@gmail.com","username":"noonedeadpunk"}},{"label":"Backport-Candidate","status":"MAY"}]}],"submit_requirements":[{"name":"Verified","description":"Verified in gate by CI","status":"SATISFIED","is_legacy":false,"submittability_expression_result":{"expression":"label:Verified\u003dMAX AND -label:Verified\u003dMIN","fulfilled":true,"status":"PASS","passing_atoms":["label:Verified\u003dMAX"],"failing_atoms":["label:Verified\u003dMIN"],"atom_explanations":{"label:Verified\u003dMAX":"","label:Verified\u003dMIN":""}}},{"name":"Backport-Candidate","description":"Backport candidate status","status":"NOT_APPLICABLE","is_legacy":false,"applicability_expression_result":{"fulfilled":false,"status":"FAIL"},"submittability_expression_result":{"expression":"is:true","fulfilled":true,"status":"NOT_EVALUATED","passing_atoms":[],"failing_atoms":[],"atom_explanations":{}}},{"name":"Code-Review","description":"Code reviewed by core reviewer","status":"SATISFIED","is_legacy":false,"submittability_expression_result":{"expression":"label:Code-Review\u003dMAX AND -label:Code-Review\u003dMIN","fulfilled":true,"status":"PASS","passing_atoms":["label:Code-Review\u003dMAX"],"failing_atoms":["label:Code-Review\u003dMIN"],"atom_explanations":{"label:Code-Review\u003dMAX":"","label:Code-Review\u003dMIN":""}}},{"name":"Workflow","description":"Approved for gate by core reviewer","status":"SATISFIED","is_legacy":false,"submittability_expression_result":{"expression":"label:Workflow\u003dMAX AND -label:Workflow\u003dMIN","fulfilled":true,"status":"PASS","passing_atoms":["label:Workflow\u003dMAX"],"failing_atoms":["label:Workflow\u003dMIN"],"atom_explanations":{"label:Workflow\u003dMAX":"","label:Workflow\u003dMIN":""}}}]}
