)]}'
{"id":"openstack%2Fopenstack-ansible-lxc_hosts~990808","triplet_id":"openstack%2Fopenstack-ansible-lxc_hosts~master~I57333775c70d0a7df24fa8b3367ebed20b5d0afe","project":"openstack/openstack-ansible-lxc_hosts","branch":"master","attention_set":{},"removed_from_attention_set":{"37598":{"account":{"_account_id":37598,"name":"Ivan Anfimov","display_name":"Ivan Anfimov","email":"lazekteam@gmail.com","username":"anfimovir"},"last_update":"2026-06-03 12:03:00.000000000","reason":"Change was submitted"}},"hashtags":[],"change_id":"I57333775c70d0a7df24fa8b3367ebed20b5d0afe","subject":"Remove redundant vars line","status":"MERGED","created":"2026-05-30 20:53:02.000000000","updated":"2026-06-03 12:04:51.000000000","submitted":"2026-06-03 12:03:00.000000000","submitter":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"total_comment_count":1,"unresolved_comment_count":0,"has_review_started":true,"submission_id":"990808","meta_rev_id":"c87ae6048e527c6d1dc1ece87e6224c8abf4e210","_number":990808,"virtual_id_number":990808,"owner":{"_account_id":37598,"name":"Ivan Anfimov","display_name":"Ivan Anfimov","email":"lazekteam@gmail.com","username":"anfimovir"},"actions":{},"labels":{"Verified":{"approved":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"all":[{"value":0,"_account_id":32666,"name":"Damian Dąbrowski","email":"damian@dabrowski.cloud","username":"ddabrowski"},{"value":0,"_account_id":28619,"name":"Dmitriy Rabotyagov","email":"noonedeadpunk@gmail.com","username":"noonedeadpunk"},{"tag":"autogenerated:zuul:gate","value":2,"date":"2026-06-03 12:03:00.000000000","permitted_voting_range":{"min":2,"max":2},"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]}],"values":{"-2":"Fails","-1":"Doesn\u0027t seem to work"," 0":"No score","+1":"Works for me","+2":"Verified"},"description":"","default_value":0,"optional":true},"Code-Review":{"approved":{"_account_id":32666,"name":"Damian Dąbrowski","email":"damian@dabrowski.cloud","username":"ddabrowski"},"all":[{"value":2,"date":"2026-06-03 09:21:00.000000000","permitted_voting_range":{"min":2,"max":2},"_account_id":32666,"name":"Damian Dąbrowski","email":"damian@dabrowski.cloud","username":"ddabrowski"},{"value":2,"date":"2026-06-02 15:03:47.000000000","permitted_voting_range":{"min":2,"max":2},"_account_id":28619,"name":"Dmitriy Rabotyagov","email":"noonedeadpunk@gmail.com","username":"noonedeadpunk"},{"value":0,"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]}],"values":{"-2":"Do not merge","-1":"This patch needs further work before it can be merged"," 0":"No score","+1":"Looks good to me, but someone else must approve","+2":"Looks good to me (core reviewer)"},"description":"","default_value":0,"optional":true},"Workflow":{"approved":{"_account_id":32666,"name":"Damian Dąbrowski","email":"damian@dabrowski.cloud","username":"ddabrowski"},"all":[{"value":1,"date":"2026-06-03 09:21:00.000000000","permitted_voting_range":{"min":1,"max":1},"_account_id":32666,"name":"Damian Dąbrowski","email":"damian@dabrowski.cloud","username":"ddabrowski"},{"value":0,"_account_id":28619,"name":"Dmitriy Rabotyagov","email":"noonedeadpunk@gmail.com","username":"noonedeadpunk"},{"value":0,"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]}],"values":{"-1":"Work in progress"," 0":"Ready for reviews","+1":"Approved"},"description":"","default_value":0,"optional":true},"Backport-Candidate":{"all":[{"value":0,"_account_id":32666,"name":"Damian Dąbrowski","email":"damian@dabrowski.cloud","username":"ddabrowski"},{"value":0,"_account_id":28619,"name":"Dmitriy Rabotyagov","email":"noonedeadpunk@gmail.com","username":"noonedeadpunk"},{"value":0,"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]}],"values":{"-1":"Not A Backport Candidate"," 0":"Backport Review Needed","+1":"Proposed Backport"},"description":"","default_value":0,"optional":true}},"removable_reviewers":[],"reviewers":{"REVIEWER":[{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},{"_account_id":28619,"name":"Dmitriy Rabotyagov","email":"noonedeadpunk@gmail.com","username":"noonedeadpunk"},{"_account_id":32666,"name":"Damian Dąbrowski","email":"damian@dabrowski.cloud","username":"ddabrowski"}]},"pending_reviewers":{},"reviewer_updates":[{"updated":"2026-05-31 00:12:19.000000000","updated_by":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"reviewer":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"state":"REVIEWER"},{"updated":"2026-06-02 15:03:47.000000000","updated_by":{"_account_id":28619,"name":"Dmitriy Rabotyagov","email":"noonedeadpunk@gmail.com","username":"noonedeadpunk"},"reviewer":{"_account_id":28619,"name":"Dmitriy Rabotyagov","email":"noonedeadpunk@gmail.com","username":"noonedeadpunk"},"state":"REVIEWER"},{"updated":"2026-06-03 09:21:00.000000000","updated_by":{"_account_id":32666,"name":"Damian Dąbrowski","email":"damian@dabrowski.cloud","username":"ddabrowski"},"reviewer":{"_account_id":32666,"name":"Damian Dąbrowski","email":"damian@dabrowski.cloud","username":"ddabrowski"},"state":"REVIEWER"}],"messages":[{"id":"43a1248b15233aeb1be1e08d5f0b58cb415c2d43","tag":"autogenerated:gerrit:newWipPatchSet","author":{"_account_id":37598,"name":"Ivan Anfimov","display_name":"Ivan Anfimov","email":"lazekteam@gmail.com","username":"anfimovir"},"date":"2026-05-30 20:53:02.000000000","message":"Uploaded patch set 1.","accounts_in_message":[],"_revision_number":1},{"id":"9959289228cf15bc26e6cf33aff5647f1caa30d8","tag":"autogenerated:gerrit:newWipPatchSet","author":{"_account_id":37598,"name":"Ivan Anfimov","display_name":"Ivan Anfimov","email":"lazekteam@gmail.com","username":"anfimovir"},"date":"2026-05-30 20:54:05.000000000","message":"Patch Set 2: Published edit on patch set 1.","accounts_in_message":[],"_revision_number":2},{"id":"a9337d11b230363985601b3c3e652acb7b534b92","author":{"_account_id":37598,"name":"Ivan Anfimov","display_name":"Ivan Anfimov","email":"lazekteam@gmail.com","username":"anfimovir"},"date":"2026-05-30 20:54:08.000000000","message":"Patch Set 2:\n\nThis change is ready for review.","accounts_in_message":[],"_revision_number":2},{"id":"78d7822f8ca29244f7563e43b24b7b653244de59","tag":"autogenerated:zuul:check","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2026-05-31 00:12:19.000000000","message":"Patch Set 2: Verified-1\n\nBuild failed (check pipeline).  For information on how to proceed, see\nhttps://docs.opendev.org/opendev/infra-manual/latest/developers.html#automated-testing\nand https://docs.openstack.org/project-team-guide/testing.html#how-to-handle-test-failures\n\nhttps://zuul.opendev.org/t/openstack/buildset/93b8a633ed9c4b4ab4b103b5699b96f2\n\n- openstack-ansible-linters-ubuntu-noble https://zuul.opendev.org/t/openstack/build/fd5bcd844ebc420eaa81f143c8dc0e31 : SUCCESS in 8m 14s\n- openstack-tox-docs https://zuul.opendev.org/t/openstack/build/3c18475967b142ab8d09f356fd57d5bd : SUCCESS in 5m 54s\n- build-openstack-releasenotes https://zuul.opendev.org/t/openstack/build/3c8bcc21def74c23981695d638cf21c9 : SUCCESS in 3m 18s\n- openstack-ansible-deploy-infra_lxc-centos-10-stream https://zuul.opendev.org/t/openstack/build/63662b1503eb4537ba6b5bbbea3e8c36 : FAILURE in 1h 52m 42s\n- openstack-ansible-deploy-infra_lxc-debian-bookworm https://zuul.opendev.org/t/openstack/build/8ffc402ba5fc4f73a1214dbcc3bf5b06 : SUCCESS in 1h 48m 51s\n- openstack-ansible-deploy-infra_lxc-debian-trixie https://zuul.opendev.org/t/openstack/build/12055f3e4ffa4cc28df3e1e8e6389c37 : SUCCESS in 1h 49m 22s\n- openstack-ansible-deploy-infra_lxc-rockylinux-10 https://zuul.opendev.org/t/openstack/build/edf26c6a2e094fd0ab5e38eee43d441d : SUCCESS in 1h 35m 35s\n- openstack-ansible-deploy-infra_lxc_tls-rockylinux-10 https://zuul.opendev.org/t/openstack/build/7eb742a080164d73b278cabb9ed25bb0 : FAILURE in 1h 34m 56s\n- openstack-ansible-deploy-infra_lxc-ubuntu-noble https://zuul.opendev.org/t/openstack/build/fadd3185bd5e4046b59ea1d62e135ef3 : SUCCESS in 1h 52m 13s\n- openstack-ansible-deploy-infra_lxc_stepca-ubuntu-noble https://zuul.opendev.org/t/openstack/build/814e72475b184f9389ea0d9648922646 : SUCCESS in 1h 52m 45s\n- openstack-ansible-deploy-infra_lxc_tls-ubuntu-noble https://zuul.opendev.org/t/openstack/build/ddd2016b2837487d8c6b8c7b49ffe0e9 : SUCCESS in 2h 28m 02s\n- openstack-ansible-upgrade-infra_lxc-ubuntu-noble https://zuul.opendev.org/t/openstack/build/347164a5da424d158a942692a91b705c : SUCCESS in 3h 08m 54s\n- openstack-ansible-deploy-hosts_lxc_zfs-ubuntu-noble https://zuul.opendev.org/t/openstack/build/ab2197d30849463ab34d6a9c109e10fb : SUCCESS in 24m 31s\n- openstack-ansible-deploy-hosts_lxc_btrfs-ubuntu-noble https://zuul.opendev.org/t/openstack/build/f5358c7d55ec4eb5970ac7259d8ba0ce : SUCCESS in 20m 49s\n- openstack-ansible-deploy-hosts_lxc_overlayfs-ubuntu-noble https://zuul.opendev.org/t/openstack/build/bcc69c95a1f646dfa50d730bf8c5995f : SUCCESS in 24m 54s\n- openstack-ansible-deploy-hosts_lxc_btrfs-debian-bookworm https://zuul.opendev.org/t/openstack/build/8ccb23057672410288146e05d91a7443 : SUCCESS in 28m 20s\n- openstack-ansible-deploy-hosts_lxc_overlayfs-debian-bookworm https://zuul.opendev.org/t/openstack/build/eb92a5fe863d44cdb70d6de366544a65 : SUCCESS in 26m 10s\n- openstack-ansible-deploy-hosts_lxc_overlayfs-rockylinux-10 https://zuul.opendev.org/t/openstack/build/2478cf814564447992765dceee56fbb2 : SUCCESS in 21m 31s\n- openstack-ansible-deploy-hosts_lxc_btrfs-debian-trixie https://zuul.opendev.org/t/openstack/build/8fb1edfa461a4292982af79d1a774929 : SUCCESS in 24m 26s\n- openstack-ansible-deploy-hosts_lxc_overlayfs-debian-trixie https://zuul.opendev.org/t/openstack/build/41154df13e1c46249b0301ec038a3dcb : SUCCESS in 25m 08s","accounts_in_message":[],"_revision_number":2},{"id":"70d134dce6e12b49ede143cf81e35ca03334e9cb","author":{"_account_id":28619,"name":"Dmitriy Rabotyagov","email":"noonedeadpunk@gmail.com","username":"noonedeadpunk"},"date":"2026-06-02 15:03:47.000000000","message":"Patch Set 2: Code-Review+2\n\n(1 comment)","accounts_in_message":[],"_revision_number":2},{"id":"f94ed28f761623d87bad54531bd47578d96a4078","tag":"autogenerated:zuul:check","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2026-06-02 17:43:40.000000000","message":"Patch Set 2: Verified+1\n\nBuild succeeded (check pipeline).\nhttps://zuul.opendev.org/t/openstack/buildset/5f61faecb34a493bba776bf80d13698c\n\n- openstack-ansible-linters-ubuntu-noble https://zuul.opendev.org/t/openstack/build/66481b5a9d5041a8bb0638c0875c75bb : SUCCESS in 14m 49s\n- openstack-tox-docs https://zuul.opendev.org/t/openstack/build/9946be8b4f1846b68a47518742343854 : SUCCESS in 7m 12s\n- build-openstack-releasenotes https://zuul.opendev.org/t/openstack/build/e0e941503f5748ff89c176fa04c7927e : SUCCESS in 2m 49s\n- openstack-ansible-deploy-infra_lxc-centos-10-stream https://zuul.opendev.org/t/openstack/build/176dc4c5fe084fefb5ea1b3e537032eb : SUCCESS in 1h 01m 42s\n- openstack-ansible-deploy-infra_lxc-debian-bookworm https://zuul.opendev.org/t/openstack/build/345f8399f5244b31ae4620a20f011815 : SUCCESS in 1h 38m 55s\n- openstack-ansible-deploy-infra_lxc-debian-trixie https://zuul.opendev.org/t/openstack/build/04b34360475a42b5909c615117ff3740 : SUCCESS in 1h 33m 45s\n- openstack-ansible-deploy-infra_lxc-rockylinux-10 https://zuul.opendev.org/t/openstack/build/d561b9edfc66415d9ce826b4142f9b47 : SUCCESS in 1h 06m 28s\n- openstack-ansible-deploy-infra_lxc_tls-rockylinux-10 https://zuul.opendev.org/t/openstack/build/e55c6119ddad44b3b987be8b21d3bb30 : SUCCESS in 1h 08m 13s\n- openstack-ansible-deploy-infra_lxc-ubuntu-noble https://zuul.opendev.org/t/openstack/build/3931e2d3b4044c3b8cef16c5a376243a : SUCCESS in 59m 23s\n- openstack-ansible-deploy-infra_lxc_stepca-ubuntu-noble https://zuul.opendev.org/t/openstack/build/76d1916de856467d9646f169dfbef0cd : SUCCESS in 1h 21m 57s\n- openstack-ansible-deploy-infra_lxc_tls-ubuntu-noble https://zuul.opendev.org/t/openstack/build/d4f5ea40b61b49049d822e338992ecff : SUCCESS in 1h 43m 05s\n- openstack-ansible-upgrade-infra_lxc-ubuntu-noble https://zuul.opendev.org/t/openstack/build/8cb41dccbd094fc9a37335af140e2a63 : SUCCESS in 2h 23m 33s\n- openstack-ansible-deploy-hosts_lxc_zfs-ubuntu-noble https://zuul.opendev.org/t/openstack/build/1ce6e945b0fb4413857a196aebfc5787 : SUCCESS in 29m 23s\n- openstack-ansible-deploy-hosts_lxc_btrfs-ubuntu-noble https://zuul.opendev.org/t/openstack/build/e25c1a28e85e4ad38012d21fbc26568a : SUCCESS in 27m 57s\n- openstack-ansible-deploy-hosts_lxc_overlayfs-ubuntu-noble https://zuul.opendev.org/t/openstack/build/290f9deaab2b42678991d3fe23bb55d2 : SUCCESS in 29m 53s\n- openstack-ansible-deploy-hosts_lxc_btrfs-debian-bookworm https://zuul.opendev.org/t/openstack/build/c8dbbbb19d8f4803850a68efb0837a72 : SUCCESS in 29m 00s\n- openstack-ansible-deploy-hosts_lxc_overlayfs-debian-bookworm https://zuul.opendev.org/t/openstack/build/53e0afaae5664c42a4551e6abe3efffa : SUCCESS in 23m 12s\n- openstack-ansible-deploy-hosts_lxc_overlayfs-rockylinux-10 https://zuul.opendev.org/t/openstack/build/54fffddf98b94bc9b4a79e50d937603b : SUCCESS in 21m 20s\n- openstack-ansible-deploy-hosts_lxc_btrfs-debian-trixie https://zuul.opendev.org/t/openstack/build/3ceceac4741446af85702a9fef92b5f1 : SUCCESS in 29m 44s\n- openstack-ansible-deploy-hosts_lxc_overlayfs-debian-trixie https://zuul.opendev.org/t/openstack/build/3fa5c2173e3a4e0e9932029f02c8fa92 : SUCCESS in 29m 06s","accounts_in_message":[],"_revision_number":2},{"id":"134a8d74c3e6769de28458ea98ba0078ab91b6e1","author":{"_account_id":32666,"name":"Damian Dąbrowski","email":"damian@dabrowski.cloud","username":"ddabrowski"},"date":"2026-06-03 09:21:00.000000000","message":"Patch Set 2: Code-Review+2 Workflow+1","accounts_in_message":[],"_revision_number":2},{"id":"96ef77d15921368fbac0a6efec7b63d8fb0e1dde","tag":"autogenerated:zuul:gate","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2026-06-03 09:27:07.000000000","message":"Patch Set 2: -Verified\n\nStarting gate jobs.","accounts_in_message":[],"_revision_number":2},{"id":"105b3a16a675b0476b3bc0f34be7352d876cbd35","tag":"autogenerated:zuul:gate","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2026-06-03 12:03:00.000000000","message":"Patch Set 2: Verified+2\n\nBuild succeeded (gate pipeline).\nhttps://zuul.opendev.org/t/openstack/buildset/7f02d065efec49f89864d3df02d0d0f9\n\n- openstack-ansible-linters-ubuntu-noble https://zuul.opendev.org/t/openstack/build/2ce579721fa0415aad6fcb09d510700d : SUCCESS in 13m 43s\n- openstack-tox-docs https://zuul.opendev.org/t/openstack/build/1ead29dcff274e15b4aeabca2e096862 : SUCCESS in 6m 25s\n- build-openstack-releasenotes https://zuul.opendev.org/t/openstack/build/4c74218ccd9c44e8895fa83e7be12c01 : SUCCESS in 5m 00s\n- openstack-ansible-deploy-infra_lxc-centos-10-stream https://zuul.opendev.org/t/openstack/build/df0225f52d3e4597b15a5168487a5fec : SUCCESS in 1h 35m 23s\n- openstack-ansible-deploy-infra_lxc-debian-bookworm https://zuul.opendev.org/t/openstack/build/1ea86fd27dbb4139a8e422f4510bb0e1 : SUCCESS in 1h 33m 33s\n- openstack-ansible-deploy-infra_lxc-debian-trixie https://zuul.opendev.org/t/openstack/build/d5c732b0c7594f9e93e70610b3a951ef : SUCCESS in 1h 42m 25s\n- openstack-ansible-deploy-infra_lxc-rockylinux-10 https://zuul.opendev.org/t/openstack/build/345324f2740942dbae240ac58134ce47 : SUCCESS in 1h 22m 29s\n- openstack-ansible-deploy-infra_lxc_tls-rockylinux-10 https://zuul.opendev.org/t/openstack/build/a2b4b00a87244f4e9e3090d6fb709468 : SUCCESS in 1h 28m 32s\n- openstack-ansible-deploy-infra_lxc-ubuntu-noble https://zuul.opendev.org/t/openstack/build/fc611acf195644559abb5c4885d0871b : SUCCESS in 1h 42m 27s\n- openstack-ansible-deploy-infra_lxc_stepca-ubuntu-noble https://zuul.opendev.org/t/openstack/build/89f62070a7d34e4eaeb188287cc777b8 : SUCCESS in 1h 41m 20s\n- openstack-ansible-deploy-infra_lxc_tls-ubuntu-noble https://zuul.opendev.org/t/openstack/build/27c2fd7afcc54e52b2fcc66f649b6955 : SUCCESS in 1h 40m 57s\n- openstack-ansible-upgrade-infra_lxc-ubuntu-noble https://zuul.opendev.org/t/openstack/build/6c3a82cb17824f13bae6033ba4e9c417 : SUCCESS in 2h 23m 15s\n- openstack-ansible-deploy-hosts_lxc_zfs-ubuntu-noble https://zuul.opendev.org/t/openstack/build/46cb6e1a51da470db76a97c7e595ea88 : SUCCESS in 29m 22s\n- openstack-ansible-deploy-hosts_lxc_btrfs-ubuntu-noble https://zuul.opendev.org/t/openstack/build/66cb74dbda1840a093fe238c33276aa2 : SUCCESS in 23m 34s\n- openstack-ansible-deploy-hosts_lxc_overlayfs-ubuntu-noble https://zuul.opendev.org/t/openstack/build/0580e16652644989aff3df7f37e69bcc : SUCCESS in 33m 47s\n- openstack-ansible-deploy-hosts_lxc_btrfs-debian-bookworm https://zuul.opendev.org/t/openstack/build/151ffb474059489cbb175c8e1c75a6fa : SUCCESS in 31m 46s\n- openstack-ansible-deploy-hosts_lxc_overlayfs-debian-bookworm https://zuul.opendev.org/t/openstack/build/f32a472dad15414582ccd9b776cb6f04 : SUCCESS in 23m 42s\n- openstack-ansible-deploy-hosts_lxc_btrfs-debian-trixie https://zuul.opendev.org/t/openstack/build/73dbe95c595f4691bfa3279669841451 : SUCCESS in 30m 53s\n- openstack-ansible-deploy-hosts_lxc_overlayfs-debian-trixie https://zuul.opendev.org/t/openstack/build/d6b1068b5a4345cb839c58231d409edf : SUCCESS in 29m 13s\n- openstack-ansible-deploy-hosts_lxc_overlayfs-rockylinux-10 https://zuul.opendev.org/t/openstack/build/b1721505ec9a4dcdb978f5bed362eb2a : SUCCESS in 25m 54s","accounts_in_message":[],"_revision_number":2},{"id":"9b93c5f572d8d13da4c630c90531ddc38a0901c8","tag":"autogenerated:gerrit:merged","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2026-06-03 12:03:00.000000000","message":"Change has been successfully merged","accounts_in_message":[],"_revision_number":2},{"id":"c87ae6048e527c6d1dc1ece87e6224c8abf4e210","tag":"autogenerated:zuul:promote","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2026-06-03 12:04:51.000000000","message":"Patch Set 2:\n\nBuild succeeded (promote pipeline).\nhttps://zuul.opendev.org/t/openstack/buildset/af3a28e0bffc47b8840126988d1e2003\n\n- promote-openstack-tox-docs https://zuul.opendev.org/t/openstack/build/bbfedbfcdc2a483687f2028024d7e676 : SUCCESS in 1m 02s","accounts_in_message":[],"_revision_number":2}],"current_revision_number":2,"current_revision":"61f1e37e94e611f0dbbaf5fcad0f8acf0b46581d","revisions":{"f7f0eb7fafe692ac5dd42a838b77ce01189e898b":{"kind":"REWORK","_number":1,"created":"2026-05-30 20:53:02.000000000","uploader":{"_account_id":37598,"name":"Ivan Anfimov","display_name":"Ivan Anfimov","email":"lazekteam@gmail.com","username":"anfimovir"},"ref":"refs/changes/08/990808/1","fetch":{"anonymous http":{"url":"https://review.opendev.org/openstack/openstack-ansible-lxc_hosts","ref":"refs/changes/08/990808/1","commands":{"Checkout":"git fetch https://review.opendev.org/openstack/openstack-ansible-lxc_hosts refs/changes/08/990808/1 \u0026\u0026 git checkout FETCH_HEAD","Cherry Pick":"git fetch https://review.opendev.org/openstack/openstack-ansible-lxc_hosts refs/changes/08/990808/1 \u0026\u0026 git cherry-pick FETCH_HEAD","Format Patch":"git fetch https://review.opendev.org/openstack/openstack-ansible-lxc_hosts refs/changes/08/990808/1 \u0026\u0026 git format-patch -1 --stdout FETCH_HEAD","Pull":"git pull https://review.opendev.org/openstack/openstack-ansible-lxc_hosts refs/changes/08/990808/1"}}},"commit":{"parents":[{"commit":"3eab0d7262eb0c79acf305a89ab3e8ebd9c4fe54","subject":"Merge \"Add support for Ubuntu 26.04\"","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/openstack-ansible-lxc_hosts/commit/3eab0d7262eb0c79acf305a89ab3e8ebd9c4fe54"}]}],"author":{"name":"Ivan Anfimov","email":"lazekteam@gmail.com","date":"2026-05-30 20:53:02.000000000","tz":0},"committer":{"name":"Ivan Anfimov","email":"lazekteam@gmail.com","date":"2026-05-30 20:53:02.000000000","tz":0},"subject":"Remove redundant vars line","message":"Remove redundant vars line\n\nChange-Id: I57333775c70d0a7df24fa8b3367ebed20b5d0afe\nSigned-off-by: Ivan Anfimov \u003clazekteam@gmail.com\u003e\n","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/openstack-ansible-lxc_hosts/commit/f7f0eb7fafe692ac5dd42a838b77ce01189e898b"}],"resolve_conflicts_web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/openstack-ansible-lxc_hosts/commit/f7f0eb7fafe692ac5dd42a838b77ce01189e898b"}]},"branch":"refs/heads/master","conflicts":{"contains_conflicts":false}},"61f1e37e94e611f0dbbaf5fcad0f8acf0b46581d":{"kind":"REWORK","_number":2,"created":"2026-05-30 20:54:05.000000000","uploader":{"_account_id":37598,"name":"Ivan Anfimov","display_name":"Ivan Anfimov","email":"lazekteam@gmail.com","username":"anfimovir"},"ref":"refs/changes/08/990808/2","fetch":{"anonymous http":{"url":"https://review.opendev.org/openstack/openstack-ansible-lxc_hosts","ref":"refs/changes/08/990808/2","commands":{"Checkout":"git fetch https://review.opendev.org/openstack/openstack-ansible-lxc_hosts refs/changes/08/990808/2 \u0026\u0026 git checkout FETCH_HEAD","Cherry Pick":"git fetch https://review.opendev.org/openstack/openstack-ansible-lxc_hosts refs/changes/08/990808/2 \u0026\u0026 git cherry-pick FETCH_HEAD","Format Patch":"git fetch https://review.opendev.org/openstack/openstack-ansible-lxc_hosts refs/changes/08/990808/2 \u0026\u0026 git format-patch -1 --stdout FETCH_HEAD","Pull":"git pull https://review.opendev.org/openstack/openstack-ansible-lxc_hosts refs/changes/08/990808/2"}}},"commit":{"parents":[{"commit":"3eab0d7262eb0c79acf305a89ab3e8ebd9c4fe54","subject":"Merge \"Add support for Ubuntu 26.04\"","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/openstack-ansible-lxc_hosts/commit/3eab0d7262eb0c79acf305a89ab3e8ebd9c4fe54"}]}],"author":{"name":"Ivan Anfimov","email":"lazekteam@gmail.com","date":"2026-05-30 20:53:02.000000000","tz":0},"committer":{"name":"Ivan Anfimov","email":"lazekteam@gmail.com","date":"2026-05-30 20:54:05.000000000","tz":0},"subject":"Remove redundant vars line","message":"Remove redundant vars line\n\nChange-Id: I57333775c70d0a7df24fa8b3367ebed20b5d0afe\nSigned-off-by: Ivan Anfimov \u003clazekteam@gmail.com\u003e\n","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/openstack-ansible-lxc_hosts/commit/61f1e37e94e611f0dbbaf5fcad0f8acf0b46581d"}],"resolve_conflicts_web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/openstack-ansible-lxc_hosts/commit/61f1e37e94e611f0dbbaf5fcad0f8acf0b46581d"}]},"branch":"refs/heads/master"}},"requirements":[],"submit_records":[{"rule_name":"gerrit~DefaultSubmitRule","status":"CLOSED","labels":[{"label":"Verified","status":"MAY","applied_by":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]}},{"label":"Code-Review","status":"MAY","applied_by":{"_account_id":28619,"name":"Dmitriy Rabotyagov","email":"noonedeadpunk@gmail.com","username":"noonedeadpunk"}},{"label":"Workflow","status":"MAY","applied_by":{"_account_id":32666,"name":"Damian Dąbrowski","email":"damian@dabrowski.cloud","username":"ddabrowski"}},{"label":"Backport-Candidate","status":"MAY"}]}],"submit_requirements":[{"name":"Verified","description":"Verified in gate by CI","status":"SATISFIED","is_legacy":false,"submittability_expression_result":{"expression":"label:Verified\u003dMAX AND -label:Verified\u003dMIN","fulfilled":true,"status":"PASS","passing_atoms":["label:Verified\u003dMAX"],"failing_atoms":["label:Verified\u003dMIN"],"atom_explanations":{"label:Verified\u003dMAX":"","label:Verified\u003dMIN":""}}},{"name":"Backport-Candidate","description":"Backport candidate status","status":"NOT_APPLICABLE","is_legacy":false,"applicability_expression_result":{"fulfilled":false,"status":"FAIL"},"submittability_expression_result":{"expression":"is:true","fulfilled":true,"status":"NOT_EVALUATED","passing_atoms":[],"failing_atoms":[],"atom_explanations":{}}},{"name":"Code-Review","description":"Code reviewed by core reviewer","status":"SATISFIED","is_legacy":false,"submittability_expression_result":{"expression":"label:Code-Review\u003dMAX AND -label:Code-Review\u003dMIN","fulfilled":true,"status":"PASS","passing_atoms":["label:Code-Review\u003dMAX"],"failing_atoms":["label:Code-Review\u003dMIN"],"atom_explanations":{"label:Code-Review\u003dMAX":"","label:Code-Review\u003dMIN":""}}},{"name":"Workflow","description":"Approved for gate by core reviewer","status":"SATISFIED","is_legacy":false,"submittability_expression_result":{"expression":"label:Workflow\u003dMAX AND -label:Workflow\u003dMIN","fulfilled":true,"status":"PASS","passing_atoms":["label:Workflow\u003dMAX"],"failing_atoms":["label:Workflow\u003dMIN"],"atom_explanations":{"label:Workflow\u003dMAX":"","label:Workflow\u003dMIN":""}}}]}
