)]}'
{"id":"openstack%2Fopenstack-ansible-memcached_server~1006918","triplet_id":"openstack%2Fopenstack-ansible-memcached_server~master~I658f493740b274bd19dcc13ca3c7db4f1ca7127c","project":"openstack/openstack-ansible-memcached_server","branch":"master","topic":"osa/molecule","attention_set":{},"removed_from_attention_set":{"28619":{"account":{"_account_id":28619,"name":"Dmitriy Rabotyagov","email":"noonedeadpunk@gmail.com","username":"noonedeadpunk"},"last_update":"2026-09-24 17:32:17.000000000","reason":"Change was submitted"}},"hashtags":[],"change_id":"I658f493740b274bd19dcc13ca3c7db4f1ca7127c","subject":"Address ansible-galaxy fetching issues.","status":"MERGED","created":"2026-09-23 09:55:53.000000000","updated":"2026-09-24 17:33:31.000000000","submitted":"2026-09-24 17:32:17.000000000","submitter":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"total_comment_count":0,"unresolved_comment_count":0,"has_review_started":true,"submission_id":"1006918-osa/molecule","meta_rev_id":"279d1a76fd5a318e34ef28628885f5308e2cbf80","_number":1006918,"virtual_id_number":1006918,"owner":{"_account_id":28619,"name":"Dmitriy Rabotyagov","email":"noonedeadpunk@gmail.com","username":"noonedeadpunk"},"actions":{},"labels":{"Verified":{"approved":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"all":[{"tag":"autogenerated:zuul:gate","value":2,"date":"2026-09-24 17:32:17.000000000","permitted_voting_range":{"min":2,"max":2},"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},{"value":0,"_account_id":37598,"name":"Ivan Anfimov","display_name":"Ivan Anfimov","email":"lazekteam@gmail.com","username":"anfimovir"},{"value":0,"_account_id":25023,"name":"Jonathan Rosser","email":"jonathan.rosser@rd.bbc.co.uk","username":"jrosser"}],"values":{"-2":"Fails","-1":"Doesn\u0027t seem to work"," 0":"No score","+1":"Works for me","+2":"Verified"},"description":"","default_value":0,"optional":true},"Code-Review":{"approved":{"_account_id":37598,"name":"Ivan Anfimov","display_name":"Ivan Anfimov","email":"lazekteam@gmail.com","username":"anfimovir"},"all":[{"value":0,"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},{"value":2,"date":"2026-09-24 01:28:56.000000000","permitted_voting_range":{"min":2,"max":2},"_account_id":37598,"name":"Ivan Anfimov","display_name":"Ivan Anfimov","email":"lazekteam@gmail.com","username":"anfimovir"},{"value":2,"date":"2026-09-24 15:20:19.000000000","permitted_voting_range":{"min":2,"max":2},"_account_id":25023,"name":"Jonathan Rosser","email":"jonathan.rosser@rd.bbc.co.uk","username":"jrosser"}],"values":{"-2":"Do not merge","-1":"This patch needs further work before it can be merged"," 0":"No score","+1":"Looks good to me, but someone else must approve","+2":"Looks good to me (core reviewer)"},"description":"","default_value":0,"optional":true},"Workflow":{"approved":{"_account_id":25023,"name":"Jonathan Rosser","email":"jonathan.rosser@rd.bbc.co.uk","username":"jrosser"},"all":[{"value":0,"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},{"value":0,"_account_id":37598,"name":"Ivan Anfimov","display_name":"Ivan Anfimov","email":"lazekteam@gmail.com","username":"anfimovir"},{"value":1,"date":"2026-09-24 15:20:19.000000000","permitted_voting_range":{"min":1,"max":1},"_account_id":25023,"name":"Jonathan Rosser","email":"jonathan.rosser@rd.bbc.co.uk","username":"jrosser"}],"values":{"-1":"Work in progress"," 0":"Ready for reviews","+1":"Approved"},"description":"","default_value":0,"optional":true},"Backport-Candidate":{"all":[{"value":0,"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},{"value":0,"_account_id":37598,"name":"Ivan Anfimov","display_name":"Ivan Anfimov","email":"lazekteam@gmail.com","username":"anfimovir"},{"value":0,"_account_id":25023,"name":"Jonathan Rosser","email":"jonathan.rosser@rd.bbc.co.uk","username":"jrosser"}],"values":{"-1":"Not A Backport Candidate"," 0":"Backport Review Needed","+1":"Proposed Backport"},"description":"","default_value":0,"optional":true}},"removable_reviewers":[],"reviewers":{"REVIEWER":[{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},{"_account_id":25023,"name":"Jonathan Rosser","email":"jonathan.rosser@rd.bbc.co.uk","username":"jrosser"},{"_account_id":37598,"name":"Ivan Anfimov","display_name":"Ivan Anfimov","email":"lazekteam@gmail.com","username":"anfimovir"}]},"pending_reviewers":{},"reviewer_updates":[{"updated":"2026-09-23 12:59:43.000000000","updated_by":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"reviewer":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"state":"REVIEWER"},{"updated":"2026-09-24 01:28:56.000000000","updated_by":{"_account_id":37598,"name":"Ivan Anfimov","display_name":"Ivan Anfimov","email":"lazekteam@gmail.com","username":"anfimovir"},"reviewer":{"_account_id":37598,"name":"Ivan Anfimov","display_name":"Ivan Anfimov","email":"lazekteam@gmail.com","username":"anfimovir"},"state":"REVIEWER"},{"updated":"2026-09-24 15:20:19.000000000","updated_by":{"_account_id":25023,"name":"Jonathan Rosser","email":"jonathan.rosser@rd.bbc.co.uk","username":"jrosser"},"reviewer":{"_account_id":25023,"name":"Jonathan Rosser","email":"jonathan.rosser@rd.bbc.co.uk","username":"jrosser"},"state":"REVIEWER"}],"messages":[{"id":"8474e57935f78211f68b70b67730f62a6c3170b1","tag":"autogenerated:gerrit:newPatchSet","author":{"_account_id":28619,"name":"Dmitriy Rabotyagov","email":"noonedeadpunk@gmail.com","username":"noonedeadpunk"},"date":"2026-09-23 09:55:53.000000000","message":"Uploaded patch set 1.","accounts_in_message":[],"_revision_number":1},{"id":"5e782109d2f7c3f2c0c329b86b48d27266b72122","tag":"autogenerated:zuul:check","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2026-09-23 12:59:43.000000000","message":"Patch Set 1: Verified+1\n\nBuild succeeded (check pipeline).\nhttps://zuul.opendev.org/t/openstack/buildset/138be40e44f6490d8f233b279a859969\n\n- openstack-ansible-tox-molecule-debian-trixie https://zuul.opendev.org/t/openstack/build/73cc42b0783d4e29baa0f69687389218 : SUCCESS in 5m 48s\n- openstack-ansible-tox-molecule-ubuntu-noble https://zuul.opendev.org/t/openstack/build/9995254d2e8744869bc81cba52511f09 : SUCCESS in 7m 40s\n- openstack-ansible-tox-molecule-ubuntu-resolute https://zuul.opendev.org/t/openstack/build/498a75a244a545708f46ab3a4b4fa1b1 : SUCCESS in 9m 04s\n- openstack-ansible-tox-molecule-centos-10-stream https://zuul.opendev.org/t/openstack/build/8f0b000a66684aa39588607dcd5956f0 : SUCCESS in 6m 57s\n- openstack-ansible-tox-molecule-rockylinux-10 https://zuul.opendev.org/t/openstack/build/1ed8dfaa713b40e0a627dc2b6dd0372e : RETRY_LIMIT in 2m 03s (non-voting)\n- openstack-ansible-deploy-infra_lxc-centos-10-stream https://zuul.opendev.org/t/openstack/build/7bb61bae32274be98ce8639677b28f46 : SUCCESS in 1h 34m 54s\n- openstack-ansible-deploy-infra_lxc-debian-trixie https://zuul.opendev.org/t/openstack/build/94d33dd1b3a94423aa4dced5a21382dd : SUCCESS in 1h 37m 36s\n- openstack-ansible-deploy-infra_lxc-rockylinux-10 https://zuul.opendev.org/t/openstack/build/086195c4a1a743fbaf558fa5ce3dd87a : SUCCESS in 1h 33m 24s\n- openstack-ansible-deploy-infra_lxc_tls-rockylinux-10 https://zuul.opendev.org/t/openstack/build/991d88eeb7154c6f8f7af00b40082d62 : SUCCESS in 1h 36m 12s\n- openstack-ansible-deploy-infra_lxc-ubuntu-noble https://zuul.opendev.org/t/openstack/build/cc7382f507de4531962de3bbda2b3667 : SUCCESS in 1h 37m 02s\n- openstack-ansible-deploy-infra_lxc-ubuntu-resolute https://zuul.opendev.org/t/openstack/build/d7d29e2b96234be18b300318000510c4 : SUCCESS in 1h 43m 44s\n- openstack-ansible-deploy-infra_lxc_stepca-ubuntu-noble https://zuul.opendev.org/t/openstack/build/3fc005410b4c4d539dfc28426f16886f : SUCCESS in 1h 24m 06s\n- openstack-ansible-deploy-infra_lxc_tls-ubuntu-noble https://zuul.opendev.org/t/openstack/build/a28c7f3994de4669892a79ff78719bd1 : SUCCESS in 1h 22m 36s\n- openstack-ansible-upgrade-infra_lxc-ubuntu-noble https://zuul.opendev.org/t/openstack/build/1ae9c05d56ae4266a5d7b8618fd60c27 : SUCCESS in 2h 13m 11s\n- openstack-ansible-linters-ubuntu-noble https://zuul.opendev.org/t/openstack/build/f90b2dea4b7b428aa3e4f4b0646271e5 : SUCCESS in 15m 23s\n- openstack-tox-docs https://zuul.opendev.org/t/openstack/build/2691f2cb8a354b9b92575ff24779af08 : SUCCESS in 6m 57s\n- build-openstack-releasenotes https://zuul.opendev.org/t/openstack/build/1ce1422e8cb043a5b1366e9039fa2637 : SUCCESS in 3m 44s","accounts_in_message":[],"_revision_number":1},{"id":"ef639e1aab5c47932c876ba02a988c0a7f552239","author":{"_account_id":37598,"name":"Ivan Anfimov","display_name":"Ivan Anfimov","email":"lazekteam@gmail.com","username":"anfimovir"},"date":"2026-09-24 01:28:56.000000000","message":"Patch Set 1: Code-Review+2","accounts_in_message":[],"_revision_number":1},{"id":"4b62426bbb68df88a29dadc3db1a90e368c99ef6","author":{"_account_id":25023,"name":"Jonathan Rosser","email":"jonathan.rosser@rd.bbc.co.uk","username":"jrosser"},"date":"2026-09-24 15:20:19.000000000","message":"Patch Set 1: Code-Review+2 Workflow+1","accounts_in_message":[],"_revision_number":1},{"id":"13e4e97bcca395e6a6cb23b1e6e223f16d864e31","tag":"autogenerated:zuul:gate","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2026-09-24 15:26:37.000000000","message":"Patch Set 1: -Verified\n\nStarting gate jobs.","accounts_in_message":[],"_revision_number":1},{"id":"c27b33b057bd1e0dabde9fab43e80c52e2756a39","tag":"autogenerated:zuul:gate","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2026-09-24 17:32:17.000000000","message":"Patch Set 1: Verified+2\n\nBuild succeeded (gate pipeline).\nhttps://zuul.opendev.org/t/openstack/buildset/165f4113147747e2a3e4e9808b68a92b\n\n- openstack-ansible-tox-molecule-debian-trixie https://zuul.opendev.org/t/openstack/build/f9d536d7f5a44b718444ece872e03ac7 : SUCCESS in 4m 47s\n- openstack-ansible-tox-molecule-ubuntu-noble https://zuul.opendev.org/t/openstack/build/8c804f98c0a6412d8c7c57c2334f1c59 : SUCCESS in 5m 25s\n- openstack-ansible-tox-molecule-ubuntu-resolute https://zuul.opendev.org/t/openstack/build/ed33e922c6354f5681467d886e55e4c1 : SUCCESS in 6m 28s\n- openstack-ansible-tox-molecule-centos-10-stream https://zuul.opendev.org/t/openstack/build/d694f0b9eb1d488da7cb290921613a79 : SUCCESS in 6m 09s\n- openstack-ansible-deploy-infra_lxc-centos-10-stream https://zuul.opendev.org/t/openstack/build/a92ccb465b814e4ab408a6c2e9286321 : SUCCESS in 1h 05m 08s\n- openstack-ansible-deploy-infra_lxc-debian-trixie https://zuul.opendev.org/t/openstack/build/bc1d8edffa45482d9a6fa6b7b147d124 : SUCCESS in 1h 00m 29s\n- openstack-ansible-deploy-infra_lxc-rockylinux-10 https://zuul.opendev.org/t/openstack/build/d8bb2a03389b4fb79ae69edc455dabac : SUCCESS in 1h 10m 59s\n- openstack-ansible-deploy-infra_lxc_tls-rockylinux-10 https://zuul.opendev.org/t/openstack/build/12486f88f5884f1a994569fbed7ee657 : SUCCESS in 1h 12m 27s\n- openstack-ansible-deploy-infra_lxc-ubuntu-noble https://zuul.opendev.org/t/openstack/build/b3f706d4a97f4dbda9ab655d564d3658 : SUCCESS in 1h 48m 29s\n- openstack-ansible-deploy-infra_lxc-ubuntu-resolute https://zuul.opendev.org/t/openstack/build/428444369d1d4a329e8e165ba24d4403 : SUCCESS in 1h 22m 50s\n- openstack-ansible-deploy-infra_lxc_stepca-ubuntu-noble https://zuul.opendev.org/t/openstack/build/a1fd6601d65b49d9b7bc11502b508e94 : SUCCESS in 1h 05m 29s\n- openstack-ansible-deploy-infra_lxc_tls-ubuntu-noble https://zuul.opendev.org/t/openstack/build/d41c562eedce44108f3e017eba0b091a : SUCCESS in 1h 44m 47s\n- openstack-ansible-upgrade-infra_lxc-ubuntu-noble https://zuul.opendev.org/t/openstack/build/c82bdd9477034147be7b2be5a84978f3 : SUCCESS in 1h 50m 46s\n- openstack-ansible-linters-ubuntu-noble https://zuul.opendev.org/t/openstack/build/256c490378b84b759976d246c8421ffe : SUCCESS in 17m 42s\n- openstack-tox-docs https://zuul.opendev.org/t/openstack/build/54b277428edc48c981fdba5307daaea7 : SUCCESS in 6m 28s\n- build-openstack-releasenotes https://zuul.opendev.org/t/openstack/build/8ddf8f0180c84a57a78a1ac1ee5018a3 : SUCCESS in 4m 02s","accounts_in_message":[],"_revision_number":1},{"id":"7a61e92ec59be232cd9c6c6eefe496ab7a91e906","tag":"autogenerated:gerrit:merged","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2026-09-24 17:32:17.000000000","message":"Change has been successfully merged","accounts_in_message":[],"_revision_number":1},{"id":"279d1a76fd5a318e34ef28628885f5308e2cbf80","tag":"autogenerated:zuul:promote","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2026-09-24 17:33:31.000000000","message":"Patch Set 1:\n\nBuild succeeded (promote pipeline).\nhttps://zuul.opendev.org/t/openstack/buildset/5ff8da02166e4ff6bc07792a9a855be4\n\n- promote-openstack-tox-docs https://zuul.opendev.org/t/openstack/build/a7b1c6979c794945a669bc1b9b1c1010 : SUCCESS in 57s","accounts_in_message":[],"_revision_number":1}],"current_revision_number":1,"current_revision":"e23aecb9edd8f0385ae8fef821d7e336e2eb60ed","revisions":{"e23aecb9edd8f0385ae8fef821d7e336e2eb60ed":{"kind":"REWORK","_number":1,"created":"2026-09-23 09:55:53.000000000","uploader":{"_account_id":28619,"name":"Dmitriy Rabotyagov","email":"noonedeadpunk@gmail.com","username":"noonedeadpunk"},"ref":"refs/changes/18/1006918/1","fetch":{"anonymous http":{"url":"https://review.opendev.org/openstack/openstack-ansible-memcached_server","ref":"refs/changes/18/1006918/1","commands":{"Checkout":"git fetch https://review.opendev.org/openstack/openstack-ansible-memcached_server refs/changes/18/1006918/1 \u0026\u0026 git checkout FETCH_HEAD","Cherry Pick":"git fetch https://review.opendev.org/openstack/openstack-ansible-memcached_server refs/changes/18/1006918/1 \u0026\u0026 git cherry-pick FETCH_HEAD","Format Patch":"git fetch https://review.opendev.org/openstack/openstack-ansible-memcached_server refs/changes/18/1006918/1 \u0026\u0026 git format-patch -1 --stdout FETCH_HEAD","Pull":"git pull https://review.opendev.org/openstack/openstack-ansible-memcached_server refs/changes/18/1006918/1"}}},"commit":{"parents":[{"commit":"5c392a9fdc1958dd56f4fa74070b9a111e673c70","subject":"Merge \"Update ansible_managed variable\"","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/openstack-ansible-memcached_server/commit/5c392a9fdc1958dd56f4fa74070b9a111e673c70"}]}],"author":{"name":"Dmitriy Rabotyagov","email":"noonedeadpunk@gmail.com","date":"2026-09-23 09:55:39.000000000","tz":120},"committer":{"name":"Dmitriy Rabotyagov","email":"noonedeadpunk@gmail.com","date":"2026-09-23 09:55:39.000000000","tz":120},"subject":"Address ansible-galaxy fetching issues.","message":"Address ansible-galaxy fetching issues.\n\nIntermittent issues reaching Ansible\u0027s galaxy for molecule\ndependencies. This can be worked around by creating\ntests/requirements.yml, which is parsed by default with\nansible-compat, and it does proceed with installation first.\nIf all dependencies are already present, molecule won\u0027t attempt\nreaching galaxy. We see a lot of failures recently on CentOS 10\nStream.\n\nChange-Id: I658f493740b274bd19dcc13ca3c7db4f1ca7127c\nSigned-off-by: Dmitriy Rabotyagov \u003cnoonedeadpunk@gmail.com\u003e\n","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/openstack-ansible-memcached_server/commit/e23aecb9edd8f0385ae8fef821d7e336e2eb60ed"}],"resolve_conflicts_web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/openstack-ansible-memcached_server/commit/e23aecb9edd8f0385ae8fef821d7e336e2eb60ed"}]},"branch":"refs/heads/master"}},"requirements":[],"submit_records":[{"rule_name":"gerrit~DefaultSubmitRule","status":"CLOSED","labels":[{"label":"Verified","status":"MAY","applied_by":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]}},{"label":"Code-Review","status":"MAY","applied_by":{"_account_id":25023,"name":"Jonathan Rosser","email":"jonathan.rosser@rd.bbc.co.uk","username":"jrosser"}},{"label":"Workflow","status":"MAY","applied_by":{"_account_id":25023,"name":"Jonathan Rosser","email":"jonathan.rosser@rd.bbc.co.uk","username":"jrosser"}},{"label":"Backport-Candidate","status":"MAY"}]}],"submit_requirements":[{"name":"Verified","description":"Verified in gate by CI","status":"SATISFIED","is_legacy":false,"submittability_expression_result":{"expression":"label:Verified\u003dMAX AND -label:Verified\u003dMIN","fulfilled":true,"status":"PASS","passing_atoms":["label:Verified\u003dMAX"],"failing_atoms":["label:Verified\u003dMIN"],"atom_explanations":{"label:Verified\u003dMAX":"","label:Verified\u003dMIN":""}}},{"name":"Backport-Candidate","description":"Backport candidate status","status":"NOT_APPLICABLE","is_legacy":false,"applicability_expression_result":{"fulfilled":false,"status":"FAIL"},"submittability_expression_result":{"expression":"is:true","fulfilled":true,"status":"NOT_EVALUATED","passing_atoms":[],"failing_atoms":[],"atom_explanations":{}}},{"name":"Code-Review","description":"Code reviewed by core reviewer","status":"SATISFIED","is_legacy":false,"submittability_expression_result":{"expression":"label:Code-Review\u003dMAX AND -label:Code-Review\u003dMIN","fulfilled":true,"status":"PASS","passing_atoms":["label:Code-Review\u003dMAX"],"failing_atoms":["label:Code-Review\u003dMIN"],"atom_explanations":{"label:Code-Review\u003dMAX":"","label:Code-Review\u003dMIN":""}}},{"name":"Workflow","description":"Approved for gate by core reviewer","status":"SATISFIED","is_legacy":false,"submittability_expression_result":{"expression":"label:Workflow\u003dMAX AND -label:Workflow\u003dMIN","fulfilled":true,"status":"PASS","passing_atoms":["label:Workflow\u003dMAX"],"failing_atoms":["label:Workflow\u003dMIN"],"atom_explanations":{"label:Workflow\u003dMAX":"","label:Workflow\u003dMIN":""}}}]}
