)]}'
{"id":"openstack%2Fopenstack-ansible-memcached_server~998690","triplet_id":"openstack%2Fopenstack-ansible-memcached_server~master~I2028a4995d277e5c52b9fd5717a1b240bdacc3fe","project":"openstack/openstack-ansible-memcached_server","branch":"master","topic":"osa/ansible_managed","attention_set":{},"removed_from_attention_set":{"28619":{"account":{"_account_id":28619,"name":"Dmitriy Rabotyagov","email":"noonedeadpunk@gmail.com","username":"noonedeadpunk"},"last_update":"2026-08-26 09:26:09.000000000","reason":"Change was submitted"},"37598":{"account":{"_account_id":37598,"name":"Ivan Anfimov","display_name":"Ivan Anfimov","email":"lazekteam@gmail.com","username":"anfimovir"},"last_update":"2026-08-26 09:26:09.000000000","reason":"Change was submitted"}},"hashtags":[],"change_id":"I2028a4995d277e5c52b9fd5717a1b240bdacc3fe","subject":"Update ansible_managed variable","status":"MERGED","created":"2026-07-25 05:50:31.000000000","updated":"2026-08-26 09:27:09.000000000","submitted":"2026-08-26 09:26:09.000000000","submitter":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"total_comment_count":1,"unresolved_comment_count":0,"has_review_started":true,"submission_id":"998690-osa/ansible_managed","meta_rev_id":"2a928bbbed62efbbd19befd7ddd35ee6ee17fdc8","_number":998690,"virtual_id_number":998690,"owner":{"_account_id":37598,"name":"Ivan Anfimov","display_name":"Ivan Anfimov","email":"lazekteam@gmail.com","username":"anfimovir"},"actions":{},"labels":{"Verified":{"approved":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"all":[{"tag":"autogenerated:zuul:gate","value":2,"date":"2026-08-26 09:26:09.000000000","permitted_voting_range":{"min":2,"max":2},"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},{"value":0,"_account_id":25023,"name":"Jonathan Rosser","email":"jonathan.rosser@rd.bbc.co.uk","username":"jrosser"}],"values":{"-2":"Fails","-1":"Doesn\u0027t seem to work"," 0":"No score","+1":"Works for me","+2":"Verified"},"description":"","default_value":0,"optional":true},"Code-Review":{"approved":{"_account_id":25023,"name":"Jonathan Rosser","email":"jonathan.rosser@rd.bbc.co.uk","username":"jrosser"},"all":[{"value":0,"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},{"value":2,"date":"2026-08-26 06:51:19.000000000","permitted_voting_range":{"min":2,"max":2},"_account_id":25023,"name":"Jonathan Rosser","email":"jonathan.rosser@rd.bbc.co.uk","username":"jrosser"}],"values":{"-2":"Do not merge","-1":"This patch needs further work before it can be merged"," 0":"No score","+1":"Looks good to me, but someone else must approve","+2":"Looks good to me (core reviewer)"},"description":"","default_value":0,"optional":true},"Workflow":{"approved":{"_account_id":25023,"name":"Jonathan Rosser","email":"jonathan.rosser@rd.bbc.co.uk","username":"jrosser"},"all":[{"value":0,"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},{"value":1,"date":"2026-08-26 06:51:19.000000000","permitted_voting_range":{"min":1,"max":1},"_account_id":25023,"name":"Jonathan Rosser","email":"jonathan.rosser@rd.bbc.co.uk","username":"jrosser"}],"values":{"-1":"Work in progress"," 0":"Ready for reviews","+1":"Approved"},"description":"","default_value":0,"optional":true},"Backport-Candidate":{"all":[{"value":0,"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},{"value":0,"_account_id":25023,"name":"Jonathan Rosser","email":"jonathan.rosser@rd.bbc.co.uk","username":"jrosser"}],"values":{"-1":"Not A Backport Candidate"," 0":"Backport Review Needed","+1":"Proposed Backport"},"description":"","default_value":0,"optional":true}},"removable_reviewers":[{"_account_id":28619,"name":"Dmitriy Rabotyagov","email":"noonedeadpunk@gmail.com","username":"noonedeadpunk"}],"reviewers":{"REVIEWER":[{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},{"_account_id":25023,"name":"Jonathan Rosser","email":"jonathan.rosser@rd.bbc.co.uk","username":"jrosser"},{"_account_id":28619,"name":"Dmitriy Rabotyagov","email":"noonedeadpunk@gmail.com","username":"noonedeadpunk"}]},"pending_reviewers":{},"reviewer_updates":[{"updated":"2026-07-25 09:18:30.000000000","updated_by":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"reviewer":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"state":"REVIEWER"},{"updated":"2026-08-20 11:11:01.000000000","updated_by":{"_account_id":28619,"name":"Dmitriy Rabotyagov","email":"noonedeadpunk@gmail.com","username":"noonedeadpunk"},"reviewer":{"_account_id":28619,"name":"Dmitriy Rabotyagov","email":"noonedeadpunk@gmail.com","username":"noonedeadpunk"},"state":"CC"},{"updated":"2026-08-24 10:32:05.000000000","updated_by":{"_account_id":37598,"name":"Ivan Anfimov","display_name":"Ivan Anfimov","email":"lazekteam@gmail.com","username":"anfimovir"},"reviewer":{"_account_id":28619,"name":"Dmitriy Rabotyagov","email":"noonedeadpunk@gmail.com","username":"noonedeadpunk"},"state":"REVIEWER"},{"updated":"2026-08-26 06:51:19.000000000","updated_by":{"_account_id":25023,"name":"Jonathan Rosser","email":"jonathan.rosser@rd.bbc.co.uk","username":"jrosser"},"reviewer":{"_account_id":25023,"name":"Jonathan Rosser","email":"jonathan.rosser@rd.bbc.co.uk","username":"jrosser"},"state":"REVIEWER"}],"messages":[{"id":"09411b388fef436a8091e9836e349d2d2bc45b31","tag":"autogenerated:gerrit:newWipPatchSet","author":{"_account_id":37598,"name":"Ivan Anfimov","display_name":"Ivan Anfimov","email":"lazekteam@gmail.com","username":"anfimovir"},"date":"2026-07-25 05:50:31.000000000","message":"Uploaded patch set 1.","accounts_in_message":[],"_revision_number":1},{"id":"b19ea8af39d1edb32e094af77b51d4407031f14b","tag":"autogenerated:gerrit:setTopic","author":{"_account_id":37598,"name":"Ivan Anfimov","display_name":"Ivan Anfimov","email":"lazekteam@gmail.com","username":"anfimovir"},"date":"2026-07-25 05:50:55.000000000","message":"Topic set to osa/ansible_managed","accounts_in_message":[],"_revision_number":1},{"id":"2964667e6e6ec2ba627c58aa15ad51585a385fdf","tag":"autogenerated:gerrit:newWipPatchSet","author":{"_account_id":37598,"name":"Ivan Anfimov","display_name":"Ivan Anfimov","email":"lazekteam@gmail.com","username":"anfimovir"},"date":"2026-07-25 05:52:12.000000000","message":"Patch Set 2: Published edit on patch set 1.","accounts_in_message":[],"_revision_number":2},{"id":"c14f897325b0ddaebd0dd390c6626bf32fd95db6","author":{"_account_id":37598,"name":"Ivan Anfimov","display_name":"Ivan Anfimov","email":"lazekteam@gmail.com","username":"anfimovir"},"date":"2026-07-25 05:52:15.000000000","message":"Patch Set 2:\n\nThis change is ready for review.","accounts_in_message":[],"_revision_number":2},{"id":"cc7fddfe4cfc977f8f2e691e7184595ee01f7448","tag":"autogenerated:zuul:check","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2026-07-25 09:18:30.000000000","message":"Patch Set 2: Verified-1\n\nBuild failed (check pipeline).  For information on how to proceed, see\nhttps://docs.opendev.org/opendev/infra-manual/latest/developers.html#automated-testing\nand https://docs.openstack.org/project-team-guide/testing.html#how-to-handle-test-failures\n\nhttps://zuul.opendev.org/t/openstack/buildset/ca5e6c1da22a4d1d86fd9b5b57bffce3\n\n- openstack-ansible-tox-molecule-debian-trixie https://zuul.opendev.org/t/openstack/build/ef3d212044814d3b990c7d9e1bdd67e3 : SUCCESS in 6m 22s\n- openstack-ansible-tox-molecule-ubuntu-noble https://zuul.opendev.org/t/openstack/build/02bc232109b4432f9e16002296f24326 : SUCCESS in 8m 39s\n- openstack-ansible-tox-molecule-ubuntu-resolute https://zuul.opendev.org/t/openstack/build/18d9d8228b4041759da7e0b95c445aaf : SUCCESS in 6m 25s\n- openstack-ansible-tox-molecule-centos-10-stream https://zuul.opendev.org/t/openstack/build/a19c2667fc874388a4b9a5a6ce1fadeb : SUCCESS in 5m 05s\n- openstack-ansible-tox-molecule-rockylinux-10 https://zuul.opendev.org/t/openstack/build/d5782b7c31e24e0bbffc2a95e9818057 : FAILURE in 4m 45s (non-voting)\n- openstack-ansible-deploy-infra_lxc-centos-10-stream https://zuul.opendev.org/t/openstack/build/e1aec8598faf4c29bebcce816c0d178b : TIMED_OUT in 3h 18m 28s\n- openstack-ansible-deploy-infra_lxc-debian-trixie https://zuul.opendev.org/t/openstack/build/1ee341930ec34b20ad9860954d4f9934 : SUCCESS in 1h 31m 17s\n- openstack-ansible-deploy-infra_lxc-rockylinux-10 https://zuul.opendev.org/t/openstack/build/c720ab3b967f4d80a9d5993eebe58d0c : SUCCESS in 1h 43m 15s\n- openstack-ansible-deploy-infra_lxc_tls-rockylinux-10 https://zuul.opendev.org/t/openstack/build/6a34bf7d6b2941f88c48da2885adafae : SUCCESS in 1h 44m 02s\n- openstack-ansible-deploy-infra_lxc-ubuntu-noble https://zuul.opendev.org/t/openstack/build/a5b0f729e2dc4e9c8828cb5242350554 : SUCCESS in 1h 31m 30s\n- openstack-ansible-deploy-infra_lxc-ubuntu-resolute https://zuul.opendev.org/t/openstack/build/2ef071c5247d4d88b91cf06588b495ad : SUCCESS in 1h 31m 00s\n- openstack-ansible-deploy-infra_lxc_stepca-ubuntu-noble https://zuul.opendev.org/t/openstack/build/bd681731f0054c70b141a7e6e8e81dc3 : SUCCESS in 1h 58m 14s\n- openstack-ansible-deploy-infra_lxc_tls-ubuntu-noble https://zuul.opendev.org/t/openstack/build/0fbbc45c73e94b65a3ac637379764e79 : SUCCESS in 1h 34m 40s\n- openstack-ansible-upgrade-infra_lxc-ubuntu-noble https://zuul.opendev.org/t/openstack/build/e4e7e9c334aa4f70ad7c06c5b2400322 : SUCCESS in 2h 03m 48s\n- openstack-ansible-linters-ubuntu-noble https://zuul.opendev.org/t/openstack/build/ce440e183d4746918c5e2d1e6db82172 : SUCCESS in 9m 48s\n- openstack-tox-docs https://zuul.opendev.org/t/openstack/build/6af502868a2a451fab0fd0850894ea8c : SUCCESS in 5m 57s\n- build-openstack-releasenotes https://zuul.opendev.org/t/openstack/build/b263a84cb81e4f93b9aae32fb899b194 : SUCCESS in 3m 59s","accounts_in_message":[],"_revision_number":2},{"id":"b9039963c7bee89e500ca39628c8623b28109c8a","author":{"_account_id":28619,"name":"Dmitriy Rabotyagov","email":"noonedeadpunk@gmail.com","username":"noonedeadpunk"},"date":"2026-08-20 11:11:01.000000000","message":"Patch Set 2:\n\n(1 comment)","accounts_in_message":[],"_revision_number":2},{"id":"7c321d178dbe4c63d8ab1482b1d38503fc79766d","tag":"autogenerated:zuul:check","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2026-08-20 16:16:22.000000000","message":"Patch Set 2: Verified+1\n\nBuild succeeded (check pipeline).\nhttps://zuul.opendev.org/t/openstack/buildset/59e4e4bd416f43ab83d8cfbf2c1c87a3\n\n- openstack-ansible-tox-molecule-debian-trixie https://zuul.opendev.org/t/openstack/build/9b067159ba9746c492fe0082c34d3455 : SUCCESS in 6m 02s\n- openstack-ansible-tox-molecule-ubuntu-noble https://zuul.opendev.org/t/openstack/build/b88868e2760a40d98ffb63a904aefe85 : SUCCESS in 6m 27s\n- openstack-ansible-tox-molecule-ubuntu-resolute https://zuul.opendev.org/t/openstack/build/128b21c5b04540ba961457cae76f19d6 : SUCCESS in 7m 07s\n- openstack-ansible-tox-molecule-centos-10-stream https://zuul.opendev.org/t/openstack/build/16c59dbd26b540a0845b3fb085a71cba : SUCCESS in 6m 54s\n- openstack-ansible-tox-molecule-rockylinux-10 https://zuul.opendev.org/t/openstack/build/97ef3e849b1647e88abd81d1407aab48 : FAILURE in 5m 39s (non-voting)\n- openstack-ansible-deploy-infra_lxc-centos-10-stream https://zuul.opendev.org/t/openstack/build/21c2e340a5c1468da7152dc60eb40820 : SUCCESS in 1h 03m 06s\n- openstack-ansible-deploy-infra_lxc-debian-trixie https://zuul.opendev.org/t/openstack/build/b580abedb22c49068e1f88f508e688cd : SUCCESS in 1h 39m 46s\n- openstack-ansible-deploy-infra_lxc-rockylinux-10 https://zuul.opendev.org/t/openstack/build/5f841266a9ad4d4ca5f8cfe919e8024f : SUCCESS in 1h 32m 56s\n- openstack-ansible-deploy-infra_lxc_tls-rockylinux-10 https://zuul.opendev.org/t/openstack/build/97500cff8dda4da280c9058d89ad23f1 : SUCCESS in 1h 40m 27s\n- openstack-ansible-deploy-infra_lxc-ubuntu-noble https://zuul.opendev.org/t/openstack/build/a4ee1358d56c4e3795941ebe1c72f574 : SUCCESS in 1h 36m 25s\n- openstack-ansible-deploy-infra_lxc-ubuntu-resolute https://zuul.opendev.org/t/openstack/build/ce3fb29478ca4b8eaf7dda02125ece0b : SUCCESS in 1h 36m 13s\n- openstack-ansible-deploy-infra_lxc_stepca-ubuntu-noble https://zuul.opendev.org/t/openstack/build/a83be19b03404800b5a04d583320e01a : SUCCESS in 1h 36m 56s\n- openstack-ansible-deploy-infra_lxc_tls-ubuntu-noble https://zuul.opendev.org/t/openstack/build/94ac7e796cc142139a1fe8b525e70732 : SUCCESS in 1h 35m 49s\n- openstack-ansible-upgrade-infra_lxc-ubuntu-noble https://zuul.opendev.org/t/openstack/build/47ea8f4e9d094cc79a4701571ec3ab74 : SUCCESS in 2h 12m 49s\n- openstack-ansible-linters-ubuntu-noble https://zuul.opendev.org/t/openstack/build/41a76a1f8ea042c9ab9a75ef22b435bf : SUCCESS in 15m 44s\n- openstack-tox-docs https://zuul.opendev.org/t/openstack/build/6ac8c3308e3142b39fdc811c95cffe7f : SUCCESS in 7m 16s\n- build-openstack-releasenotes https://zuul.opendev.org/t/openstack/build/6cfccf76e9d6422abb0a731f76f747ed : SUCCESS in 2m 40s","accounts_in_message":[],"_revision_number":2},{"id":"94a7294f3658fb879781bd8c386b205362da8b13","author":{"_account_id":25023,"name":"Jonathan Rosser","email":"jonathan.rosser@rd.bbc.co.uk","username":"jrosser"},"date":"2026-08-26 06:51:19.000000000","message":"Patch Set 2: Code-Review+2 Workflow+1","accounts_in_message":[],"_revision_number":2},{"id":"6c0057c6685defa5f3b165b84e8423d59e7d0223","tag":"autogenerated:zuul:gate","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2026-08-26 06:53:48.000000000","message":"Patch Set 2: -Verified\n\nStarting gate jobs.","accounts_in_message":[],"_revision_number":2},{"id":"aa8603cf6caa653861cc8fb16748caf2f6f43209","tag":"autogenerated:zuul:gate","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2026-08-26 09:26:09.000000000","message":"Patch Set 2: Verified+2\n\nBuild succeeded (gate pipeline).\nhttps://zuul.opendev.org/t/openstack/buildset/63095837142243b299a58a88ac0de055\n\n- openstack-ansible-tox-molecule-debian-trixie https://zuul.opendev.org/t/openstack/build/696b887a83c04096abba6fad8cfbe2bb : SUCCESS in 3m 41s\n- openstack-ansible-tox-molecule-ubuntu-noble https://zuul.opendev.org/t/openstack/build/f5ac3164eca44d8297a749fa3d5f76f2 : SUCCESS in 6m 08s\n- openstack-ansible-tox-molecule-ubuntu-resolute https://zuul.opendev.org/t/openstack/build/cf5b8657783748cab52eabf0eb8f47cd : SUCCESS in 7m 44s\n- openstack-ansible-tox-molecule-centos-10-stream https://zuul.opendev.org/t/openstack/build/9c6927e687fe48f1ba783a849b26445c : SUCCESS in 5m 24s\n- openstack-ansible-deploy-infra_lxc-centos-10-stream https://zuul.opendev.org/t/openstack/build/a68a7aeaa1cc4e3ca5bc11cb2358dc03 : SUCCESS in 1h 00m 21s\n- openstack-ansible-deploy-infra_lxc-debian-trixie https://zuul.opendev.org/t/openstack/build/c862c1442e7c4e3ea62c4eaf0a15d4af : SUCCESS in 59m 39s\n- openstack-ansible-deploy-infra_lxc-rockylinux-10 https://zuul.opendev.org/t/openstack/build/2086496ad0e34c95824bfdb254e60efd : SUCCESS in 1h 00m 35s\n- openstack-ansible-deploy-infra_lxc_tls-rockylinux-10 https://zuul.opendev.org/t/openstack/build/8d35d6e8f1ef403795ff609b4de64b82 : SUCCESS in 1h 17m 27s\n- openstack-ansible-deploy-infra_lxc-ubuntu-noble https://zuul.opendev.org/t/openstack/build/40e0217349a142d1b1e11eff9205f240 : SUCCESS in 1h 39m 34s\n- openstack-ansible-deploy-infra_lxc-ubuntu-resolute https://zuul.opendev.org/t/openstack/build/a44c79967e5748138f0772bf153139ef : SUCCESS in 1h 43m 48s\n- openstack-ansible-deploy-infra_lxc_stepca-ubuntu-noble https://zuul.opendev.org/t/openstack/build/5baed70e275c43fe9838aa106abcc0ad : SUCCESS in 1h 08m 33s\n- openstack-ansible-deploy-infra_lxc_tls-ubuntu-noble https://zuul.opendev.org/t/openstack/build/3a85de27dc214fd5bfefcde874c7304d : SUCCESS in 1h 41m 19s\n- openstack-ansible-upgrade-infra_lxc-ubuntu-noble https://zuul.opendev.org/t/openstack/build/d8572b413d08451b863e7149a090025d : SUCCESS in 2h 20m 44s\n- openstack-ansible-linters-ubuntu-noble https://zuul.opendev.org/t/openstack/build/9ef0c162c8644973a7bffef2cfc31b7f : SUCCESS in 11m 42s\n- openstack-tox-docs https://zuul.opendev.org/t/openstack/build/75f76c03ac1b4132b4428dd8500c9335 : SUCCESS in 6m 10s\n- build-openstack-releasenotes https://zuul.opendev.org/t/openstack/build/c5b89e316b274d57bcc0a86436f6580c : SUCCESS in 3m 15s","accounts_in_message":[],"_revision_number":2},{"id":"f069b49365fbf701b5a015eccbf1916895d818a7","tag":"autogenerated:gerrit:merged","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2026-08-26 09:26:09.000000000","message":"Change has been successfully merged","accounts_in_message":[],"_revision_number":2},{"id":"2a928bbbed62efbbd19befd7ddd35ee6ee17fdc8","tag":"autogenerated:zuul:promote","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2026-08-26 09:27:09.000000000","message":"Patch Set 2:\n\nBuild succeeded (promote pipeline).\nhttps://zuul.opendev.org/t/openstack/buildset/c71efacce7344f9eaa794a5e8488c8ff\n\n- promote-openstack-tox-docs https://zuul.opendev.org/t/openstack/build/fcce624e7abb4adea665495e31c49d00 : SUCCESS in 47s","accounts_in_message":[],"_revision_number":2}],"current_revision_number":2,"current_revision":"41b88222fd30ffc070be6ec2c03b57df718c0a56","revisions":{"65c8cb273b1ca0ae2023c0d6bee790a0b8e15253":{"kind":"REWORK","_number":1,"created":"2026-07-25 05:50:31.000000000","uploader":{"_account_id":37598,"name":"Ivan Anfimov","display_name":"Ivan Anfimov","email":"lazekteam@gmail.com","username":"anfimovir"},"ref":"refs/changes/90/998690/1","fetch":{"anonymous http":{"url":"https://review.opendev.org/openstack/openstack-ansible-memcached_server","ref":"refs/changes/90/998690/1","commands":{"Checkout":"git fetch https://review.opendev.org/openstack/openstack-ansible-memcached_server refs/changes/90/998690/1 \u0026\u0026 git checkout FETCH_HEAD","Cherry Pick":"git fetch https://review.opendev.org/openstack/openstack-ansible-memcached_server refs/changes/90/998690/1 \u0026\u0026 git cherry-pick FETCH_HEAD","Format Patch":"git fetch https://review.opendev.org/openstack/openstack-ansible-memcached_server refs/changes/90/998690/1 \u0026\u0026 git format-patch -1 --stdout FETCH_HEAD","Pull":"git pull https://review.opendev.org/openstack/openstack-ansible-memcached_server refs/changes/90/998690/1"}}},"commit":{"parents":[{"commit":"a9756120811fa2b381dc788170b56136008ceb67","subject":"Merge \"Remove memcached_disable_privatedevices\"","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/openstack-ansible-memcached_server/commit/a9756120811fa2b381dc788170b56136008ceb67"}]}],"author":{"name":"Ivan Anfimov","email":"lazekteam@gmail.com","date":"2026-07-25 05:50:31.000000000","tz":0},"committer":{"name":"Ivan Anfimov","email":"lazekteam@gmail.com","date":"2026-07-25 05:50:31.000000000","tz":0},"subject":"Update ansible_managed variable","message":"Update ansible_managed variable\n\nDeprecated in Ansible Core 2.23 [0]\n\n[0] https://docs.ansible.com/projects/ansible-core/devel/reference_appendices/config.html#default-managed-str\n\nChange-Id: I2028a4995d277e5c52b9fd5717a1b240bdacc3fe\nSigned-off-by: Ivan Anfimov \u003clazekteam@gmail.com\u003e\n","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/openstack-ansible-memcached_server/commit/65c8cb273b1ca0ae2023c0d6bee790a0b8e15253"}],"resolve_conflicts_web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/openstack-ansible-memcached_server/commit/65c8cb273b1ca0ae2023c0d6bee790a0b8e15253"}]},"branch":"refs/heads/master","conflicts":{"no_base_reason":"NO_MERGE_PERFORMED","contains_conflicts":false}},"41b88222fd30ffc070be6ec2c03b57df718c0a56":{"kind":"REWORK","_number":2,"created":"2026-07-25 05:52:12.000000000","uploader":{"_account_id":37598,"name":"Ivan Anfimov","display_name":"Ivan Anfimov","email":"lazekteam@gmail.com","username":"anfimovir"},"ref":"refs/changes/90/998690/2","fetch":{"anonymous http":{"url":"https://review.opendev.org/openstack/openstack-ansible-memcached_server","ref":"refs/changes/90/998690/2","commands":{"Checkout":"git fetch https://review.opendev.org/openstack/openstack-ansible-memcached_server refs/changes/90/998690/2 \u0026\u0026 git checkout FETCH_HEAD","Cherry Pick":"git fetch https://review.opendev.org/openstack/openstack-ansible-memcached_server refs/changes/90/998690/2 \u0026\u0026 git cherry-pick FETCH_HEAD","Format Patch":"git fetch https://review.opendev.org/openstack/openstack-ansible-memcached_server refs/changes/90/998690/2 \u0026\u0026 git format-patch -1 --stdout FETCH_HEAD","Pull":"git pull https://review.opendev.org/openstack/openstack-ansible-memcached_server refs/changes/90/998690/2"}}},"commit":{"parents":[{"commit":"a9756120811fa2b381dc788170b56136008ceb67","subject":"Merge \"Remove memcached_disable_privatedevices\"","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/openstack-ansible-memcached_server/commit/a9756120811fa2b381dc788170b56136008ceb67"}]}],"author":{"name":"Ivan Anfimov","email":"lazekteam@gmail.com","date":"2026-07-25 05:50:31.000000000","tz":0},"committer":{"name":"Ivan Anfimov","email":"lazekteam@gmail.com","date":"2026-07-25 05:52:12.000000000","tz":0},"subject":"Update ansible_managed variable","message":"Update ansible_managed variable\n\nDeprecated in Ansible Core 2.23 [0]\n\n[0] https://docs.ansible.com/projects/ansible-core/devel/reference_appendices/config.html#default-managed-str\n\nChange-Id: I2028a4995d277e5c52b9fd5717a1b240bdacc3fe\nSigned-off-by: Ivan Anfimov \u003clazekteam@gmail.com\u003e\n","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/openstack-ansible-memcached_server/commit/41b88222fd30ffc070be6ec2c03b57df718c0a56"}],"resolve_conflicts_web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/openstack-ansible-memcached_server/commit/41b88222fd30ffc070be6ec2c03b57df718c0a56"}]},"branch":"refs/heads/master"}},"requirements":[],"submit_records":[{"rule_name":"gerrit~DefaultSubmitRule","status":"CLOSED","labels":[{"label":"Verified","status":"MAY","applied_by":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]}},{"label":"Code-Review","status":"MAY","applied_by":{"_account_id":25023,"name":"Jonathan Rosser","email":"jonathan.rosser@rd.bbc.co.uk","username":"jrosser"}},{"label":"Workflow","status":"MAY","applied_by":{"_account_id":25023,"name":"Jonathan Rosser","email":"jonathan.rosser@rd.bbc.co.uk","username":"jrosser"}},{"label":"Backport-Candidate","status":"MAY"}]}],"submit_requirements":[{"name":"Verified","description":"Verified in gate by CI","status":"SATISFIED","is_legacy":false,"submittability_expression_result":{"expression":"label:Verified\u003dMAX AND -label:Verified\u003dMIN","fulfilled":true,"status":"PASS","passing_atoms":["label:Verified\u003dMAX"],"failing_atoms":["label:Verified\u003dMIN"],"atom_explanations":{"label:Verified\u003dMAX":"","label:Verified\u003dMIN":""}}},{"name":"Backport-Candidate","description":"Backport candidate status","status":"NOT_APPLICABLE","is_legacy":false,"applicability_expression_result":{"fulfilled":false,"status":"FAIL"},"submittability_expression_result":{"expression":"is:true","fulfilled":true,"status":"NOT_EVALUATED","passing_atoms":[],"failing_atoms":[],"atom_explanations":{}}},{"name":"Code-Review","description":"Code reviewed by core reviewer","status":"SATISFIED","is_legacy":false,"submittability_expression_result":{"expression":"label:Code-Review\u003dMAX AND -label:Code-Review\u003dMIN","fulfilled":true,"status":"PASS","passing_atoms":["label:Code-Review\u003dMAX"],"failing_atoms":["label:Code-Review\u003dMIN"],"atom_explanations":{"label:Code-Review\u003dMAX":"","label:Code-Review\u003dMIN":""}}},{"name":"Workflow","description":"Approved for gate by core reviewer","status":"SATISFIED","is_legacy":false,"submittability_expression_result":{"expression":"label:Workflow\u003dMAX AND -label:Workflow\u003dMIN","fulfilled":true,"status":"PASS","passing_atoms":["label:Workflow\u003dMAX"],"failing_atoms":["label:Workflow\u003dMIN"],"atom_explanations":{"label:Workflow\u003dMAX":"","label:Workflow\u003dMIN":""}}}]}
