)]}'
{"id":"openstack%2Fopenstack-ansible-openstack_hosts~673631","triplet_id":"openstack%2Fopenstack-ansible-openstack_hosts~master~I7af695c45d4a556e7695efaecb3fa1f21db405df","project":"openstack/openstack-ansible-openstack_hosts","branch":"master","topic":"openstack/openstack-ansible-tests/sync-tests","hashtags":[],"change_id":"I7af695c45d4a556e7695efaecb3fa1f21db405df","subject":"Updated from OpenStack Ansible Tests","status":"MERGED","created":"2019-07-30 19:36:19.000000000","updated":"2019-07-30 20:15:17.000000000","submitted":"2019-07-30 20:15:17.000000000","submitter":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"total_comment_count":0,"unresolved_comment_count":0,"has_review_started":true,"submission_id":"673631-1564517717714-66f2565c","meta_rev_id":"6d3a9bcfe41e959149a25a6d470dee7c6550bd19","_number":673631,"virtual_id_number":673631,"owner":{"_account_id":11131,"name":"OpenStack Proposal Bot","username":"proposal-bot","tags":["SERVICE_USER"]},"actions":{},"labels":{"Verified":{"approved":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"all":[{"value":0,"_account_id":1004,"name":"Mohammed Naser","email":"mnaser@vexxhost.com","username":"mnaser"},{"value":2,"date":"2019-07-30 20:15:17.000000000","permitted_voting_range":{"min":2,"max":2},"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]}],"values":{"-2":"Fails","-1":"Doesn\u0027t seem to work"," 0":"No score","+1":"Works for me","+2":"Verified"},"description":"","default_value":0,"optional":true},"Code-Review":{"approved":{"_account_id":1004,"name":"Mohammed Naser","email":"mnaser@vexxhost.com","username":"mnaser"},"all":[{"value":2,"date":"2019-07-30 20:02:31.000000000","permitted_voting_range":{"min":2,"max":2},"_account_id":1004,"name":"Mohammed Naser","email":"mnaser@vexxhost.com","username":"mnaser"},{"value":0,"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]}],"values":{"-2":"Do not merge","-1":"This patch needs further work before it can be merged"," 0":"No score","+1":"Looks good to me, but someone else must approve","+2":"Looks good to me (core reviewer)"},"description":"","default_value":0,"optional":true},"Workflow":{"approved":{"_account_id":1004,"name":"Mohammed Naser","email":"mnaser@vexxhost.com","username":"mnaser"},"all":[{"value":1,"date":"2019-07-30 20:02:31.000000000","permitted_voting_range":{"min":1,"max":1},"_account_id":1004,"name":"Mohammed Naser","email":"mnaser@vexxhost.com","username":"mnaser"},{"value":0,"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]}],"values":{"-1":"Work in progress"," 0":"Ready for reviews","+1":"Approved"},"description":"","default_value":0,"optional":true},"Backport-Candidate":{"all":[{"value":0,"_account_id":1004,"name":"Mohammed Naser","email":"mnaser@vexxhost.com","username":"mnaser"},{"value":0,"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]}],"values":{"-1":"Not A Backport Candidate"," 0":"Backport Review Needed","+1":"Proposed Backport"},"description":"","default_value":0,"optional":true}},"removable_reviewers":[],"reviewers":{"REVIEWER":[{"_account_id":1004,"name":"Mohammed Naser","email":"mnaser@vexxhost.com","username":"mnaser"},{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]}]},"pending_reviewers":{},"reviewer_updates":[{"updated":"2019-07-30 20:02:31.000000000","updated_by":{"_account_id":1004,"name":"Mohammed Naser","email":"mnaser@vexxhost.com","username":"mnaser"},"reviewer":{"_account_id":1004,"name":"Mohammed Naser","email":"mnaser@vexxhost.com","username":"mnaser"},"state":"REVIEWER"},{"updated":"2019-07-30 20:15:17.000000000","updated_by":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"reviewer":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"state":"REVIEWER"}],"messages":[{"id":"9ceccd471dfefeaefa1a18b647675b0bb63c01ed","author":{"_account_id":11131,"name":"OpenStack Proposal Bot","username":"proposal-bot","tags":["SERVICE_USER"]},"date":"2019-07-30 19:36:19.000000000","message":"Uploaded patch set 1.","accounts_in_message":[],"_revision_number":1},{"id":"9b49655c125f87d06574ef3b390a1c89ce3eb468","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2019-07-30 19:58:44.000000000","message":"Patch Set 1: Verified+1\n\nBuild succeeded (check pipeline).\n\n- openstack-ansible-linters http://logs.openstack.org/31/673631/1/check/openstack-ansible-linters/917bac1/ : SUCCESS in 5m 49s\n- requirements-check http://logs.openstack.org/31/673631/1/check/requirements-check/1e78d23/ : SUCCESS in 2m 14s\n- openstack-tox-docs http://logs.openstack.org/31/673631/1/check/openstack-tox-docs/14787e2/html/ : SUCCESS in 3m 46s\n- build-openstack-releasenotes http://logs.openstack.org/31/673631/1/check/build-openstack-releasenotes/9ad51dd/html/ : SUCCESS in 3m 09s","accounts_in_message":[],"_revision_number":1},{"id":"e0eb93c108a5d1cb78c097fc1311c8188dacada0","author":{"_account_id":1004,"name":"Mohammed Naser","email":"mnaser@vexxhost.com","username":"mnaser"},"date":"2019-07-30 20:02:31.000000000","message":"Patch Set 1: Code-Review+2 Workflow+1","accounts_in_message":[],"_revision_number":1},{"id":"5e364b2d50b164b241fa5f8367e8bfd2a5bc60b1","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2019-07-30 20:02:42.000000000","message":"Patch Set 1: -Verified\n\nStarting gate jobs.","accounts_in_message":[],"_revision_number":1},{"id":"b1bde4d74f475a4527d91863e8e5f14073ad93e8","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2019-07-30 20:15:17.000000000","message":"Change has been successfully merged by Zuul","accounts_in_message":[],"_revision_number":1},{"id":"a746d0c7fb51512871ef8f54758ee08f527272df","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2019-07-30 20:15:17.000000000","message":"Patch Set 1: Verified+2\n\nBuild succeeded (gate pipeline).\n\n- openstack-ansible-linters http://logs.openstack.org/31/673631/1/gate/openstack-ansible-linters/cbaf76e/ : SUCCESS in 5m 30s\n- requirements-check http://logs.openstack.org/31/673631/1/gate/requirements-check/cf015ad/ : SUCCESS in 2m 15s\n- openstack-tox-docs http://logs.openstack.org/31/673631/1/gate/openstack-tox-docs/d6c454b/html/ : SUCCESS in 2m 44s\n- build-openstack-releasenotes http://logs.openstack.org/31/673631/1/gate/build-openstack-releasenotes/2299157/html/ : SUCCESS in 3m 51s","accounts_in_message":[],"_revision_number":1}],"current_revision_number":1,"current_revision":"5ec996021cd694016b98c30a19fcc710aa18696c","revisions":{"5ec996021cd694016b98c30a19fcc710aa18696c":{"kind":"REWORK","_number":1,"created":"2019-07-30 19:36:19.000000000","uploader":{"_account_id":11131,"name":"OpenStack Proposal Bot","username":"proposal-bot","tags":["SERVICE_USER"]},"ref":"refs/changes/31/673631/1","fetch":{"anonymous http":{"url":"https://review.opendev.org/openstack/openstack-ansible-openstack_hosts","ref":"refs/changes/31/673631/1","commands":{"Checkout":"git fetch https://review.opendev.org/openstack/openstack-ansible-openstack_hosts refs/changes/31/673631/1 \u0026\u0026 git checkout FETCH_HEAD","Cherry Pick":"git fetch https://review.opendev.org/openstack/openstack-ansible-openstack_hosts refs/changes/31/673631/1 \u0026\u0026 git cherry-pick FETCH_HEAD","Format Patch":"git fetch https://review.opendev.org/openstack/openstack-ansible-openstack_hosts refs/changes/31/673631/1 \u0026\u0026 git format-patch -1 --stdout FETCH_HEAD","Pull":"git pull https://review.opendev.org/openstack/openstack-ansible-openstack_hosts refs/changes/31/673631/1"}}},"commit":{"parents":[{"commit":"43f52279c650561d31209e2c63ee831adf83a379","subject":"Merge \"Add bit_length_power_of_2 to filters folder\"","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/openstack-ansible-openstack_hosts/commit/43f52279c650561d31209e2c63ee831adf83a379"}]}],"author":{"name":"OpenStack Proposal Bot","email":"openstack-infra@lists.openstack.org","date":"2019-07-30 19:36:15.000000000","tz":0},"committer":{"name":"OpenStack Proposal Bot","email":"openstack-infra@lists.openstack.org","date":"2019-07-30 19:36:15.000000000","tz":0},"subject":"Updated from OpenStack Ansible Tests","message":"Updated from OpenStack Ansible Tests\n\nChange-Id: I7af695c45d4a556e7695efaecb3fa1f21db405df\n","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/openstack-ansible-openstack_hosts/commit/5ec996021cd694016b98c30a19fcc710aa18696c"}],"resolve_conflicts_web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/openstack-ansible-openstack_hosts/commit/5ec996021cd694016b98c30a19fcc710aa18696c"}]},"branch":"refs/heads/master"}},"requirements":[],"submit_records":[],"submit_requirements":[]}
