)]}'
{"id":"openstack%2Fopenstack-ansible-openstack_hosts~758570","triplet_id":"openstack%2Fopenstack-ansible-openstack_hosts~master~I94fbc9fc75e188d3a4a334ffb08836d431003764","project":"openstack/openstack-ansible-openstack_hosts","branch":"master","hashtags":[],"change_id":"I94fbc9fc75e188d3a4a334ffb08836d431003764","subject":"Set UCA repo to Victoria","status":"MERGED","created":"2020-10-16 14:06:29.000000000","updated":"2020-10-19 15:43:29.000000000","submitted":"2020-10-19 15:39:52.000000000","submitter":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"total_comment_count":0,"unresolved_comment_count":0,"has_review_started":true,"submission_id":"758570-1603121992427-577aa814","meta_rev_id":"43b03c7a5f443548196f59b54b277f5b229a008a","_number":758570,"virtual_id_number":758570,"owner":{"_account_id":28619,"name":"Dmitriy Rabotyagov","email":"noonedeadpunk@gmail.com","username":"noonedeadpunk"},"actions":{},"labels":{"Verified":{"approved":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"all":[{"value":0,"_account_id":31542,"name":"Andrew Bonney","email":"andrew.bonney@bbc.co.uk","username":"andrewbonney"},{"value":0,"date":"2020-10-19 10:37:13.000000000","_account_id":28619,"name":"Dmitriy Rabotyagov","email":"noonedeadpunk@gmail.com","username":"noonedeadpunk"},{"tag":"autogenerated:zuul:gate","value":2,"date":"2020-10-19 15:39:52.000000000","post_submit":true,"permitted_voting_range":{"min":2,"max":2},"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},{"value":0,"_account_id":25023,"name":"Jonathan Rosser","email":"jonathan.rosser@rd.bbc.co.uk","username":"jrosser"}],"values":{"-2":"Fails","-1":"Doesn\u0027t seem to work"," 0":"No score","+1":"Works for me","+2":"Verified"},"description":"","default_value":0,"optional":true},"Code-Review":{"approved":{"_account_id":31542,"name":"Andrew Bonney","email":"andrew.bonney@bbc.co.uk","username":"andrewbonney"},"all":[{"value":2,"date":"2020-10-19 07:34:34.000000000","permitted_voting_range":{"min":2,"max":2},"_account_id":31542,"name":"Andrew Bonney","email":"andrew.bonney@bbc.co.uk","username":"andrewbonney"},{"value":0,"_account_id":28619,"name":"Dmitriy Rabotyagov","email":"noonedeadpunk@gmail.com","username":"noonedeadpunk"},{"value":0,"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},{"value":2,"date":"2020-10-16 17:36:07.000000000","permitted_voting_range":{"min":2,"max":2},"_account_id":25023,"name":"Jonathan Rosser","email":"jonathan.rosser@rd.bbc.co.uk","username":"jrosser"}],"values":{"-2":"Do not merge","-1":"This patch needs further work before it can be merged"," 0":"No score","+1":"Looks good to me, but someone else must approve","+2":"Looks good to me (core reviewer)"},"description":"","default_value":0,"optional":true},"Workflow":{"approved":{"_account_id":31542,"name":"Andrew Bonney","email":"andrew.bonney@bbc.co.uk","username":"andrewbonney"},"all":[{"value":1,"date":"2020-10-19 07:34:34.000000000","permitted_voting_range":{"min":1,"max":1},"_account_id":31542,"name":"Andrew Bonney","email":"andrew.bonney@bbc.co.uk","username":"andrewbonney"},{"value":0,"_account_id":28619,"name":"Dmitriy Rabotyagov","email":"noonedeadpunk@gmail.com","username":"noonedeadpunk"},{"value":0,"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},{"value":0,"_account_id":25023,"name":"Jonathan Rosser","email":"jonathan.rosser@rd.bbc.co.uk","username":"jrosser"}],"values":{"-1":"Work in progress"," 0":"Ready for reviews","+1":"Approved"},"description":"","default_value":0,"optional":true},"Backport-Candidate":{"all":[{"value":0,"_account_id":31542,"name":"Andrew Bonney","email":"andrew.bonney@bbc.co.uk","username":"andrewbonney"},{"value":0,"_account_id":28619,"name":"Dmitriy Rabotyagov","email":"noonedeadpunk@gmail.com","username":"noonedeadpunk"},{"value":0,"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},{"value":0,"_account_id":25023,"name":"Jonathan Rosser","email":"jonathan.rosser@rd.bbc.co.uk","username":"jrosser"}],"values":{"-1":"Not A Backport Candidate"," 0":"Backport Review Needed","+1":"Proposed Backport"},"description":"","default_value":0,"optional":true}},"removable_reviewers":[],"reviewers":{"REVIEWER":[{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},{"_account_id":25023,"name":"Jonathan Rosser","email":"jonathan.rosser@rd.bbc.co.uk","username":"jrosser"},{"_account_id":28619,"name":"Dmitriy Rabotyagov","email":"noonedeadpunk@gmail.com","username":"noonedeadpunk"},{"_account_id":31542,"name":"Andrew Bonney","email":"andrew.bonney@bbc.co.uk","username":"andrewbonney"}]},"pending_reviewers":{},"reviewer_updates":[{"updated":"2020-10-16 17:36:07.000000000","updated_by":{"_account_id":25023,"name":"Jonathan Rosser","email":"jonathan.rosser@rd.bbc.co.uk","username":"jrosser"},"reviewer":{"_account_id":25023,"name":"Jonathan Rosser","email":"jonathan.rosser@rd.bbc.co.uk","username":"jrosser"},"state":"REVIEWER"},{"updated":"2020-10-19 07:34:34.000000000","updated_by":{"_account_id":31542,"name":"Andrew Bonney","email":"andrew.bonney@bbc.co.uk","username":"andrewbonney"},"reviewer":{"_account_id":31542,"name":"Andrew Bonney","email":"andrew.bonney@bbc.co.uk","username":"andrewbonney"},"state":"REVIEWER"},{"updated":"2020-10-19 15:39:52.000000000","updated_by":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"reviewer":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"state":"REVIEWER"}],"messages":[{"id":"6b21e0b0fde88a8d50fd9b7c46a307a60eb4cca4","author":{"_account_id":28619,"name":"Dmitriy Rabotyagov","email":"noonedeadpunk@gmail.com","username":"noonedeadpunk"},"date":"2020-10-16 14:06:29.000000000","message":"Uploaded patch set 1.","accounts_in_message":[],"_revision_number":1},{"id":"fd34d5e9369e28dab87263245617718b68a2cb08","tag":"autogenerated:zuul:check","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2020-10-16 16:08:35.000000000","message":"Patch Set 1: Verified+1\n\nBuild succeeded (check pipeline).\n\n- openstack-ansible-linters https://zuul.opendev.org/t/openstack/build/3cb58cec904f4da08df5abde419b2bf1 : SUCCESS in 9m 23s\n- openstack-ansible-functional-debian-buster https://zuul.opendev.org/t/openstack/build/b42682009d00484f9d1c43916daa49d5 : SUCCESS in 13m 17s\n- openstack-ansible-functional-centos-8 https://zuul.opendev.org/t/openstack/build/f3c7d2c576b74c758c355dd0196dd6dc : SUCCESS in 16m 10s\n- openstack-ansible-functional-ubuntu-bionic https://zuul.opendev.org/t/openstack/build/d106c3ff9b674b859721b5ac126dbebb : SUCCESS in 14m 30s\n- openstack-ansible-functional-ubuntu-focal https://zuul.opendev.org/t/openstack/build/804ad68b47694847ab49e3cd9f7d0c83 : SUCCESS in 13m 59s\n- openstack-ansible-deploy-aio_metal-debian-buster https://zuul.opendev.org/t/openstack/build/4298d944a7284f91adacb1c41f3f9c5c : SUCCESS in 1h 23m 59s\n- openstack-ansible-deploy-aio_metal-centos-8 https://zuul.opendev.org/t/openstack/build/0a568eb718ce4e25a7eab60776de1372 : SUCCESS in 1h 41m 10s\n- openstack-ansible-deploy-aio_metal-ubuntu-bionic https://zuul.opendev.org/t/openstack/build/165dccf5eb17493dba1a3e3351a029ea : SUCCESS in 1h 38m 53s\n- openstack-ansible-deploy-aio_metal-ubuntu-focal https://zuul.opendev.org/t/openstack/build/8f588b9c42374649a9a8f144ace2a49f : SUCCESS in 1h 19m 48s\n- openstack-ansible-upgrade-aio_metal-centos-8 https://zuul.opendev.org/t/openstack/build/c1e780aa787348dd9f03bb193c26c35a : FAILURE in 1h 18m 18s (non-voting)\n- openstack-ansible-upgrade-aio_metal-ubuntu-focal https://zuul.opendev.org/t/openstack/build/11cea769cfae4a94bb268b3bc57313c8 : SUCCESS in 1h 56m 04s\n- openstack-ansible-deploy-aio_distro_metal-centos-8 https://zuul.opendev.org/t/openstack/build/b3718b5922ae436d81c592cf1de452d1 : SUCCESS in 1h 19m 44s\n- openstack-ansible-deploy-aio_distro_metal-ubuntu-focal https://zuul.opendev.org/t/openstack/build/08f00f9457784d7eae8dd761d55589dc : SUCCESS in 1h 02m 15s\n- openstack-tox-docs https://zuul.opendev.org/t/openstack/build/7faf2e14dc3744bdb75106fb3abf677f : SUCCESS in 7m 56s\n- build-openstack-releasenotes https://zuul.opendev.org/t/openstack/build/8ca09a804b0d4054b0ece8050bffaefd : SUCCESS in 5m 48s","accounts_in_message":[],"_revision_number":1},{"id":"0bda1bfeecb037120ab0533ee9db0a5be16a70d8","author":{"_account_id":25023,"name":"Jonathan Rosser","email":"jonathan.rosser@rd.bbc.co.uk","username":"jrosser"},"date":"2020-10-16 17:36:07.000000000","message":"Patch Set 1: Code-Review+2","accounts_in_message":[],"_revision_number":1},{"id":"ca93bcbcca51a619bd66ee9fc587262955c61690","author":{"_account_id":31542,"name":"Andrew Bonney","email":"andrew.bonney@bbc.co.uk","username":"andrewbonney"},"date":"2020-10-19 07:34:34.000000000","message":"Patch Set 1: Code-Review+2 Workflow+1","accounts_in_message":[],"_revision_number":1},{"id":"0baae6612d164d2f9a74251358be79538a0cfc51","tag":"autogenerated:zuul:gate","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2020-10-19 07:34:45.000000000","message":"Patch Set 1: -Verified\n\nStarting gate jobs.","accounts_in_message":[],"_revision_number":1},{"id":"32a1a29bb83cb2a7a27272255069fdb8c9bf7130","tag":"autogenerated:zuul:gate","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2020-10-19 10:23:22.000000000","message":"Patch Set 1: Verified-2\n\nBuild failed (gate pipeline).  For information on how to proceed, see\nhttps://docs.opendev.org/opendev/infra-manual/latest/developers.html#automated-testing\n\n\n- openstack-ansible-linters https://zuul.opendev.org/t/openstack/build/fc898d0264e74596b1d1a88021344af2 : SUCCESS in 12m 55s\n- openstack-ansible-functional-debian-buster https://zuul.opendev.org/t/openstack/build/789a57d687464cabaf32efd7954d6d30 : SUCCESS in 16m 15s\n- openstack-ansible-functional-ubuntu-bionic https://zuul.opendev.org/t/openstack/build/d94a0bb75e5e4198a0df3077d3f6bce5 : SUCCESS in 15m 57s\n- openstack-ansible-deploy-aio_metal-debian-buster https://zuul.opendev.org/t/openstack/build/b22d155c5c98409a818f444888a66320 : SUCCESS in 1h 20m 24s\n- openstack-ansible-deploy-aio_metal-centos-8 https://zuul.opendev.org/t/openstack/build/21be7ebcfa134c6eb86372d107bf592b : SUCCESS in 1h 36m 23s\n- openstack-ansible-deploy-aio_metal-ubuntu-bionic https://zuul.opendev.org/t/openstack/build/f5f182369d5b42a49c98be6571b1d930 : SUCCESS in 1h 27m 15s\n- openstack-ansible-deploy-aio_metal-ubuntu-focal https://zuul.opendev.org/t/openstack/build/97803747c8f348e98c6314aac49788d1 : FAILURE in 57m 05s\n- openstack-ansible-deploy-aio_distro_metal-centos-8 https://zuul.opendev.org/t/openstack/build/6089c9d936a246cf9497a3f4dd372a35 : SUCCESS in 1h 08m 28s\n- openstack-ansible-deploy-aio_distro_metal-ubuntu-focal https://zuul.opendev.org/t/openstack/build/9bec95f4b85b4dbb8f7875ea88061ec1 : SUCCESS in 1h 02m 13s\n- openstack-tox-docs https://zuul.opendev.org/t/openstack/build/a564539094684627a0014241de552b53 : SUCCESS in 8m 46s\n- build-openstack-releasenotes https://zuul.opendev.org/t/openstack/build/728c7502a058472c847bcb96bd17132e : SUCCESS in 7m 20s","accounts_in_message":[],"_revision_number":1},{"id":"75fd37da0ebd1075175bfdeebfba7ebcefdbdd2a","author":{"_account_id":28619,"name":"Dmitriy Rabotyagov","email":"noonedeadpunk@gmail.com","username":"noonedeadpunk"},"date":"2020-10-19 10:37:13.000000000","message":"Patch Set 1:\n\nrecheck","accounts_in_message":[],"_revision_number":1},{"id":"1476a6fc8a1eccf94912d8b6a71cbcf9d77edea7","tag":"autogenerated:zuul:check","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2020-10-19 13:38:09.000000000","message":"Patch Set 1: Verified+1\n\nBuild succeeded (check pipeline).\n\n- openstack-ansible-linters https://zuul.opendev.org/t/openstack/build/98de4ab5e133496c91f423525952f8fe : SUCCESS in 11m 46s\n- openstack-ansible-functional-debian-buster https://zuul.opendev.org/t/openstack/build/2b65b5a160f146f98caf5628ba643c50 : SUCCESS in 13m 51s\n- openstack-ansible-functional-centos-8 https://zuul.opendev.org/t/openstack/build/a6cfc31741b64aa491f292bcdac6302e : SUCCESS in 15m 09s\n- openstack-ansible-functional-ubuntu-bionic https://zuul.opendev.org/t/openstack/build/bff9d30aedb8405a9a31066d57279c25 : SUCCESS in 13m 30s\n- openstack-ansible-functional-ubuntu-focal https://zuul.opendev.org/t/openstack/build/b255b0bff2584884abaf74b5401b9c31 : SUCCESS in 15m 07s\n- openstack-ansible-deploy-aio_metal-debian-buster https://zuul.opendev.org/t/openstack/build/92e8cc41ebcc49238e9400305a6fabc4 : SUCCESS in 1h 15m 06s\n- openstack-ansible-deploy-aio_metal-centos-8 https://zuul.opendev.org/t/openstack/build/98f091ae1d2046aaa441f65163249181 : SUCCESS in 1h 27m 39s\n- openstack-ansible-deploy-aio_metal-ubuntu-bionic https://zuul.opendev.org/t/openstack/build/6d50c706c2534f2989e4bda6755795b6 : SUCCESS in 1h 08m 54s\n- openstack-ansible-deploy-aio_metal-ubuntu-focal https://zuul.opendev.org/t/openstack/build/91c8a908877247899ab42beab4639cc2 : SUCCESS in 1h 07m 18s\n- openstack-ansible-upgrade-aio_metal-centos-8 https://zuul.opendev.org/t/openstack/build/bdeb6608d0854743b0a65b613b499bb4 : SUCCESS in 2h 17m 06s (non-voting)\n- openstack-ansible-upgrade-aio_metal-ubuntu-focal https://zuul.opendev.org/t/openstack/build/390a6fc8e3614d998669eec1d8cd3d21 : SUCCESS in 1h 39m 20s\n- openstack-ansible-deploy-aio_distro_metal-centos-8 https://zuul.opendev.org/t/openstack/build/0a1b697a38fd42d4aac5a8f34cbb74ce : SUCCESS in 1h 11m 14s\n- openstack-ansible-deploy-aio_distro_metal-ubuntu-focal https://zuul.opendev.org/t/openstack/build/e094ac01fbef4625a028a0ef1dd956b0 : SUCCESS in 56m 51s\n- openstack-tox-docs https://zuul.opendev.org/t/openstack/build/924513f8be5947b8a878db596186cd24 : SUCCESS in 7m 45s\n- build-openstack-releasenotes https://zuul.opendev.org/t/openstack/build/f93ba444de034a1081925a1ba57f9527 : SUCCESS in 6m 16s","accounts_in_message":[],"_revision_number":1},{"id":"605c1b79d233812cca09c238638662d7ba45ecb2","tag":"autogenerated:zuul:gate","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2020-10-19 13:38:20.000000000","message":"Patch Set 1: -Verified\n\nStarting gate jobs.","accounts_in_message":[],"_revision_number":1},{"id":"b2fd5d82c3ca02b1654db65c5e8b0abb73c87fe6","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2020-10-19 15:39:52.000000000","message":"Change has been successfully merged by Zuul","accounts_in_message":[],"_revision_number":1},{"id":"ea8b28710ac2becbfe62e187439b19910e9d6b5d","tag":"autogenerated:zuul:gate","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2020-10-19 15:39:52.000000000","message":"Patch Set 1: Verified+2\n\nBuild succeeded (gate pipeline).\n\n- openstack-ansible-linters https://zuul.opendev.org/t/openstack/build/0ea577faf15d4037b20fd32541139414 : SUCCESS in 12m 49s\n- openstack-ansible-functional-debian-buster https://zuul.opendev.org/t/openstack/build/47fd5ecb585542d78a1dab9004989ae0 : SUCCESS in 14m 22s\n- openstack-ansible-functional-ubuntu-bionic https://zuul.opendev.org/t/openstack/build/3828b742c55d47f588b31d367c440001 : SUCCESS in 15m 02s\n- openstack-ansible-deploy-aio_metal-debian-buster https://zuul.opendev.org/t/openstack/build/04c276a1d07b45f49c1ab614c49be295 : SUCCESS in 1h 14m 15s\n- openstack-ansible-deploy-aio_metal-centos-8 https://zuul.opendev.org/t/openstack/build/c2fd0f60096f47c7964167e24c7f9129 : SUCCESS in 1h 38m 13s\n- openstack-ansible-deploy-aio_metal-ubuntu-bionic https://zuul.opendev.org/t/openstack/build/ca96080101b7492ba7ff6b39388ffca1 : SUCCESS in 1h 51m 59s\n- openstack-ansible-deploy-aio_metal-ubuntu-focal https://zuul.opendev.org/t/openstack/build/3bd9681141bf427e9b8cce7ea9ed95d7 : SUCCESS in 1h 20m 37s\n- openstack-ansible-deploy-aio_distro_metal-centos-8 https://zuul.opendev.org/t/openstack/build/5ef6f79be8934638b313df8191cd1fc8 : SUCCESS in 1h 10m 21s\n- openstack-ansible-deploy-aio_distro_metal-ubuntu-focal https://zuul.opendev.org/t/openstack/build/535a96f03ed444749a2ef5ecc2c4f02f : SUCCESS in 52m 46s\n- openstack-tox-docs https://zuul.opendev.org/t/openstack/build/ee450e0382f84a8dacfc98c4cf204421 : SUCCESS in 9m 03s\n- build-openstack-releasenotes https://zuul.opendev.org/t/openstack/build/fd4e0291bf4e4b45a5dc024d0d5ae958 : SUCCESS in 8m 13s","accounts_in_message":[],"_revision_number":1},{"id":"43b03c7a5f443548196f59b54b277f5b229a008a","tag":"autogenerated:zuul:promote","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2020-10-19 15:43:29.000000000","message":"Patch Set 1:\n\nBuild succeeded (promote pipeline).\n\n- promote-openstack-tox-docs https://zuul.opendev.org/t/openstack/build/e7991815b20648d481102499dfd4db99 : SUCCESS in 2m 47s\n- promote-openstack-releasenotes https://zuul.opendev.org/t/openstack/build/17463f48cc6c467ab7c8e91ac73a3504 : SUCCESS in 1m 39s","accounts_in_message":[],"_revision_number":1}],"current_revision_number":1,"current_revision":"d7a4ff5a6a429509d589bd2248d0f92c4c461d25","revisions":{"d7a4ff5a6a429509d589bd2248d0f92c4c461d25":{"kind":"REWORK","_number":1,"created":"2020-10-16 14:06:29.000000000","uploader":{"_account_id":28619,"name":"Dmitriy Rabotyagov","email":"noonedeadpunk@gmail.com","username":"noonedeadpunk"},"ref":"refs/changes/70/758570/1","fetch":{"anonymous http":{"url":"https://review.opendev.org/openstack/openstack-ansible-openstack_hosts","ref":"refs/changes/70/758570/1","commands":{"Checkout":"git fetch https://review.opendev.org/openstack/openstack-ansible-openstack_hosts refs/changes/70/758570/1 \u0026\u0026 git checkout FETCH_HEAD","Cherry Pick":"git fetch https://review.opendev.org/openstack/openstack-ansible-openstack_hosts refs/changes/70/758570/1 \u0026\u0026 git cherry-pick FETCH_HEAD","Format Patch":"git fetch https://review.opendev.org/openstack/openstack-ansible-openstack_hosts refs/changes/70/758570/1 \u0026\u0026 git format-patch -1 --stdout FETCH_HEAD","Pull":"git pull https://review.opendev.org/openstack/openstack-ansible-openstack_hosts refs/changes/70/758570/1"}}},"commit":{"parents":[{"commit":"15c1522eabfb05f9c8feb4438c96c3264bddaab8","subject":"Merge \"Ensure SSL certificate and key directories and symlinks are present\"","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/openstack-ansible-openstack_hosts/commit/15c1522eabfb05f9c8feb4438c96c3264bddaab8"}]}],"author":{"name":"Dmitriy Rabotyagov","email":"noonedeadpunk@ya.ru","date":"2020-10-16 14:06:19.000000000","tz":180},"committer":{"name":"Dmitriy Rabotyagov","email":"noonedeadpunk@ya.ru","date":"2020-10-16 14:06:19.000000000","tz":180},"subject":"Set UCA repo to Victoria","message":"Set UCA repo to Victoria\n\nSince UCA provides V only for 20.04, we are not able to provide\ndistro installs for 18.04\nStill we\u0027re leaving UCA fixed to U for bionic to get required KVM\nlibraries and other dependencies.\n\nChange-Id: I94fbc9fc75e188d3a4a334ffb08836d431003764\n","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/openstack-ansible-openstack_hosts/commit/d7a4ff5a6a429509d589bd2248d0f92c4c461d25"}],"resolve_conflicts_web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/openstack-ansible-openstack_hosts/commit/d7a4ff5a6a429509d589bd2248d0f92c4c461d25"}]},"branch":"refs/heads/master"}},"requirements":[],"submit_records":[],"submit_requirements":[]}
