)]}'
{"id":"openstack%2Fopenstack-ansible-openstack_hosts~970169","triplet_id":"openstack%2Fopenstack-ansible-openstack_hosts~master~I0fa3161d67a4cedffa889964261f752b0d65a999","project":"openstack/openstack-ansible-openstack_hosts","branch":"master","attention_set":{},"removed_from_attention_set":{"37632":{"account":{"_account_id":37632,"name":"Dmitriy Chubinidze","email":"dcu995@gmail.com","username":"chubinidzedr"},"last_update":"2025-12-09 08:35:22.000000000","reason":"removed on reply"},"28619":{"account":{"_account_id":28619,"name":"Dmitriy Rabotyagov","email":"noonedeadpunk@gmail.com","username":"noonedeadpunk"},"last_update":"2025-12-09 08:33:58.000000000","reason":"\u003cGERRIT_ACCOUNT_28619\u003e replied on the change","reason_account":{"_account_id":28619,"name":"Dmitriy Rabotyagov","email":"noonedeadpunk@gmail.com","username":"noonedeadpunk"}},"37598":{"account":{"_account_id":37598,"name":"Ivan Anfimov","display_name":"Ivan Anfimov","email":"lazekteam@gmail.com","username":"anfimovir"},"last_update":"2025-12-09 10:03:19.000000000","reason":"Removed by \u003cGERRIT_ACCOUNT_37598\u003e by clicking the attention icon","reason_account":{"_account_id":37598,"name":"Ivan Anfimov","display_name":"Ivan Anfimov","email":"lazekteam@gmail.com","username":"anfimovir"}},"25023":{"account":{"_account_id":25023,"name":"Jonathan Rosser","email":"jonathan.rosser@rd.bbc.co.uk","username":"jrosser"},"last_update":"2025-12-09 10:50:56.000000000","reason":"Change was submitted"}},"hashtags":[],"change_id":"I0fa3161d67a4cedffa889964261f752b0d65a999","subject":"Remove outdated configure for yum","status":"MERGED","created":"2025-12-08 18:38:15.000000000","updated":"2025-12-09 10:51:48.000000000","submitted":"2025-12-09 10:50:56.000000000","submitter":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"total_comment_count":0,"unresolved_comment_count":0,"has_review_started":true,"submission_id":"970169","meta_rev_id":"aecf0050a103bcf4113cbdfb5361a4f20e78327d","_number":970169,"virtual_id_number":970169,"owner":{"_account_id":37598,"name":"Ivan Anfimov","display_name":"Ivan Anfimov","email":"lazekteam@gmail.com","username":"anfimovir"},"actions":{},"labels":{"Verified":{"approved":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"all":[{"value":0,"_account_id":37632,"name":"Dmitriy Chubinidze","email":"dcu995@gmail.com","username":"chubinidzedr"},{"value":0,"_account_id":31542,"name":"Andrew Bonney","email":"andrew.bonney@bbc.co.uk","username":"andrewbonney"},{"value":0,"_account_id":28619,"name":"Dmitriy Rabotyagov","email":"noonedeadpunk@gmail.com","username":"noonedeadpunk"},{"tag":"autogenerated:zuul:gate","value":2,"date":"2025-12-09 10:50:56.000000000","permitted_voting_range":{"min":2,"max":2},"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]}],"values":{"-2":"Fails","-1":"Doesn\u0027t seem to work"," 0":"No score","+1":"Works for me","+2":"Verified"},"description":"","default_value":0,"optional":true},"Code-Review":{"approved":{"_account_id":31542,"name":"Andrew Bonney","email":"andrew.bonney@bbc.co.uk","username":"andrewbonney"},"recommended":{"_account_id":37632,"name":"Dmitriy Chubinidze","email":"dcu995@gmail.com","username":"chubinidzedr"},"all":[{"value":1,"date":"2025-12-09 08:35:22.000000000","permitted_voting_range":{"min":1,"max":1},"_account_id":37632,"name":"Dmitriy Chubinidze","email":"dcu995@gmail.com","username":"chubinidzedr"},{"value":2,"date":"2025-12-09 08:57:30.000000000","permitted_voting_range":{"min":2,"max":2},"_account_id":31542,"name":"Andrew Bonney","email":"andrew.bonney@bbc.co.uk","username":"andrewbonney"},{"value":2,"date":"2025-12-09 08:33:58.000000000","permitted_voting_range":{"min":2,"max":2},"_account_id":28619,"name":"Dmitriy Rabotyagov","email":"noonedeadpunk@gmail.com","username":"noonedeadpunk"},{"value":0,"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]}],"values":{"-2":"Do not merge","-1":"This patch needs further work before it can be merged"," 0":"No score","+1":"Looks good to me, but someone else must approve","+2":"Looks good to me (core reviewer)"},"description":"","value":1,"default_value":0,"optional":true},"Workflow":{"approved":{"_account_id":31542,"name":"Andrew Bonney","email":"andrew.bonney@bbc.co.uk","username":"andrewbonney"},"all":[{"value":0,"_account_id":37632,"name":"Dmitriy Chubinidze","email":"dcu995@gmail.com","username":"chubinidzedr"},{"value":1,"date":"2025-12-09 08:57:30.000000000","permitted_voting_range":{"min":1,"max":1},"_account_id":31542,"name":"Andrew Bonney","email":"andrew.bonney@bbc.co.uk","username":"andrewbonney"},{"value":0,"_account_id":28619,"name":"Dmitriy Rabotyagov","email":"noonedeadpunk@gmail.com","username":"noonedeadpunk"},{"value":0,"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]}],"values":{"-1":"Work in progress"," 0":"Ready for reviews","+1":"Approved"},"description":"","default_value":0,"optional":true},"Backport-Candidate":{"all":[{"value":0,"_account_id":37632,"name":"Dmitriy Chubinidze","email":"dcu995@gmail.com","username":"chubinidzedr"},{"value":0,"_account_id":31542,"name":"Andrew Bonney","email":"andrew.bonney@bbc.co.uk","username":"andrewbonney"},{"value":0,"_account_id":28619,"name":"Dmitriy Rabotyagov","email":"noonedeadpunk@gmail.com","username":"noonedeadpunk"},{"value":0,"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]}],"values":{"-1":"Not A Backport Candidate"," 0":"Backport Review Needed","+1":"Proposed Backport"},"description":"","default_value":0,"optional":true}},"removable_reviewers":[{"_account_id":25023,"name":"Jonathan Rosser","email":"jonathan.rosser@rd.bbc.co.uk","username":"jrosser"}],"reviewers":{"REVIEWER":[{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},{"_account_id":25023,"name":"Jonathan Rosser","email":"jonathan.rosser@rd.bbc.co.uk","username":"jrosser"},{"_account_id":28619,"name":"Dmitriy Rabotyagov","email":"noonedeadpunk@gmail.com","username":"noonedeadpunk"},{"_account_id":31542,"name":"Andrew Bonney","email":"andrew.bonney@bbc.co.uk","username":"andrewbonney"},{"_account_id":37632,"name":"Dmitriy Chubinidze","email":"dcu995@gmail.com","username":"chubinidzedr"}]},"pending_reviewers":{},"reviewer_updates":[{"updated":"2025-12-08 18:38:53.000000000","updated_by":{"_account_id":37598,"name":"Ivan Anfimov","display_name":"Ivan Anfimov","email":"lazekteam@gmail.com","username":"anfimovir"},"reviewer":{"_account_id":37632,"name":"Dmitriy Chubinidze","email":"dcu995@gmail.com","username":"chubinidzedr"},"state":"REVIEWER"},{"updated":"2025-12-08 19:36:26.000000000","updated_by":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"reviewer":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"state":"REVIEWER"},{"updated":"2025-12-08 19:43:04.000000000","updated_by":{"_account_id":37598,"name":"Ivan Anfimov","display_name":"Ivan Anfimov","email":"lazekteam@gmail.com","username":"anfimovir"},"reviewer":{"_account_id":25023,"name":"Jonathan Rosser","email":"jonathan.rosser@rd.bbc.co.uk","username":"jrosser"},"state":"REVIEWER"},{"updated":"2025-12-08 19:43:04.000000000","updated_by":{"_account_id":37598,"name":"Ivan Anfimov","display_name":"Ivan Anfimov","email":"lazekteam@gmail.com","username":"anfimovir"},"reviewer":{"_account_id":28619,"name":"Dmitriy Rabotyagov","email":"noonedeadpunk@gmail.com","username":"noonedeadpunk"},"state":"REVIEWER"},{"updated":"2025-12-09 08:57:30.000000000","updated_by":{"_account_id":31542,"name":"Andrew Bonney","email":"andrew.bonney@bbc.co.uk","username":"andrewbonney"},"reviewer":{"_account_id":31542,"name":"Andrew Bonney","email":"andrew.bonney@bbc.co.uk","username":"andrewbonney"},"state":"REVIEWER"}],"messages":[{"id":"a0f7367e89c5f3912ed9f12e5017a1416e8b62d0","tag":"autogenerated:gerrit:newWipPatchSet","author":{"_account_id":37598,"name":"Ivan Anfimov","display_name":"Ivan Anfimov","email":"lazekteam@gmail.com","username":"anfimovir"},"date":"2025-12-08 18:38:15.000000000","message":"Uploaded patch set 1.","accounts_in_message":[],"_revision_number":1},{"id":"6123e49e8826dba00df80db37e70d41ada1ad322","tag":"autogenerated:gerrit:newWipPatchSet","author":{"_account_id":37598,"name":"Ivan Anfimov","display_name":"Ivan Anfimov","email":"lazekteam@gmail.com","username":"anfimovir"},"date":"2025-12-08 18:38:43.000000000","message":"Patch Set 2: Published edit on patch set 1.","accounts_in_message":[],"_revision_number":2},{"id":"a233cb7d4119d6dc2e0dae90866f767c6a2d6e97","tag":"autogenerated:zuul:check","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2025-12-08 19:36:26.000000000","message":"Patch Set 2: Verified+1\n\nBuild succeeded (check pipeline).\nhttps://zuul.opendev.org/t/openstack/buildset/a47eea6388964799898c6de2e25f7d24\n\n- openstack-ansible-deploy-hosts_metal-centos-10-stream https://zuul.opendev.org/t/openstack/build/9399d7165b764ecc9a8d22ad461dfe77 : SUCCESS in 15m 36s\n- openstack-ansible-deploy-hosts_metal-debian-bookworm https://zuul.opendev.org/t/openstack/build/5eaa809c40a448c1bc868a2a8c4f8828 : SUCCESS in 13m 03s\n- openstack-ansible-deploy-hosts_metal-debian-trixie https://zuul.opendev.org/t/openstack/build/77666e59e6ba4b659458f309d50ea68c : SUCCESS in 7m 51s\n- openstack-ansible-deploy-hosts_metal-rockylinux-10 https://zuul.opendev.org/t/openstack/build/3cb98988e10c4b5c91b0505924198f8d : SUCCESS in 12m 31s\n- openstack-ansible-deploy-hosts_metal-ubuntu-noble https://zuul.opendev.org/t/openstack/build/f0f89cd51bc24006bcc4b2966d4ded4d : SUCCESS in 11m 22s\n- openstack-ansible-deploy-hosts_distro_lxc-rockylinux-10 https://zuul.opendev.org/t/openstack/build/1be7417df092432db75a87f9f447cd7f : FAILURE in 8m 13s (non-voting)\n- openstack-ansible-deploy-hosts_distro_lxc-ubuntu-noble https://zuul.opendev.org/t/openstack/build/d2a842f222814b66b7d950a01b8ab4cb : SUCCESS in 20m 32s\n- openstack-ansible-upgrade-infra_lxc-ubuntu-noble https://zuul.opendev.org/t/openstack/build/f0b047914bec417e9bc15fc800ca47c4 : SUCCESS in 55m 13s\n- openstack-tox-docs https://zuul.opendev.org/t/openstack/build/ac9a71b3edad476cabd5d937a996766b : SUCCESS in 6m 31s\n- openstack-ansible-linters-ubuntu-noble https://zuul.opendev.org/t/openstack/build/8c940f1fd5ca4cb0982c0231f845693d : SUCCESS in 6m 05s","accounts_in_message":[],"_revision_number":2},{"id":"5675d5785f3b6b0018263b7eef6e38310561d0aa","tag":"autogenerated:gerrit:newWipPatchSet","author":{"_account_id":37598,"name":"Ivan Anfimov","display_name":"Ivan Anfimov","email":"lazekteam@gmail.com","username":"anfimovir"},"date":"2025-12-08 19:42:56.000000000","message":"Patch Set 3: Commit message was updated.\n\nOutdated Votes:\n* Verified+1\n","accounts_in_message":[],"_revision_number":3},{"id":"19bfb31608b6181eff6c44cbb52030f8c25bba6f","author":{"_account_id":37598,"name":"Ivan Anfimov","display_name":"Ivan Anfimov","email":"lazekteam@gmail.com","username":"anfimovir"},"date":"2025-12-08 19:43:04.000000000","message":"Patch Set 3:\n\nThis change is ready for review.","accounts_in_message":[],"_revision_number":3},{"id":"253b6117bb297a13e4e296b601a3843d138619a2","tag":"autogenerated:gerrit:newPatchSet","author":{"_account_id":37598,"name":"Ivan Anfimov","display_name":"Ivan Anfimov","email":"lazekteam@gmail.com","username":"anfimovir"},"date":"2025-12-08 19:44:42.000000000","message":"Patch Set 4: Commit message was updated.","accounts_in_message":[],"_revision_number":4},{"id":"4863abf049ed1109fe70fa7f468f316ca234079b","tag":"autogenerated:zuul:check","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2025-12-08 20:41:27.000000000","message":"Patch Set 4: Verified+1\n\nBuild succeeded (check pipeline).\nhttps://zuul.opendev.org/t/openstack/buildset/1c76b96a730a4078b6ec7552b189f627\n\n- openstack-ansible-deploy-hosts_metal-centos-10-stream https://zuul.opendev.org/t/openstack/build/12eed0eca61b4468adb17c10aa81fdf3 : SUCCESS in 10m 34s\n- openstack-ansible-deploy-hosts_metal-debian-bookworm https://zuul.opendev.org/t/openstack/build/1278af6050c747cba9cb9a32bbf8266e : SUCCESS in 15m 32s\n- openstack-ansible-deploy-hosts_metal-debian-trixie https://zuul.opendev.org/t/openstack/build/7347dec3e45547e0b06a95c83021f4ab : SUCCESS in 9m 14s\n- openstack-ansible-deploy-hosts_metal-rockylinux-10 https://zuul.opendev.org/t/openstack/build/5c34b51e4a1d4a7084c71a2925c4ce63 : SUCCESS in 11m 49s\n- openstack-ansible-deploy-hosts_metal-ubuntu-noble https://zuul.opendev.org/t/openstack/build/40ce5beecefb4805acf77e7b0085338b : SUCCESS in 13m 09s\n- openstack-ansible-deploy-hosts_distro_lxc-rockylinux-10 https://zuul.opendev.org/t/openstack/build/9e122a04fe794ebc817140c4d7d34e04 : FAILURE in 11m 36s (non-voting)\n- openstack-ansible-deploy-hosts_distro_lxc-ubuntu-noble https://zuul.opendev.org/t/openstack/build/f1b94c34bc294be492f9d611d108d6c9 : SUCCESS in 23m 45s\n- openstack-ansible-upgrade-infra_lxc-ubuntu-noble https://zuul.opendev.org/t/openstack/build/a9df9907f2474de3ba41b8d37938c206 : SUCCESS in 54m 06s\n- openstack-tox-docs https://zuul.opendev.org/t/openstack/build/5e0279d5e24543abb73584ade22a63c0 : SUCCESS in 5m 50s\n- openstack-ansible-linters-ubuntu-noble https://zuul.opendev.org/t/openstack/build/009239bc767a46309d16a3bd911363ad : SUCCESS in 4m 39s","accounts_in_message":[],"_revision_number":4},{"id":"89d2cffac0d5ed7ec294859a875e40557176ae74","author":{"_account_id":28619,"name":"Dmitriy Rabotyagov","email":"noonedeadpunk@gmail.com","username":"noonedeadpunk"},"date":"2025-12-09 08:33:58.000000000","message":"Patch Set 4: Code-Review+2","accounts_in_message":[],"_revision_number":4},{"id":"9907c64c37e1cabffb3eaaca03df276d82a00d82","author":{"_account_id":37632,"name":"Dmitriy Chubinidze","email":"dcu995@gmail.com","username":"chubinidzedr"},"date":"2025-12-09 08:35:22.000000000","message":"Patch Set 4: Code-Review+1","accounts_in_message":[],"_revision_number":4},{"id":"395cb76d1c2b29ae6580e864d13ff2acf6167636","author":{"_account_id":31542,"name":"Andrew Bonney","email":"andrew.bonney@bbc.co.uk","username":"andrewbonney"},"date":"2025-12-09 08:57:30.000000000","message":"Patch Set 4: Code-Review+2 Workflow+1","accounts_in_message":[],"_revision_number":4},{"id":"ee887e684eee599df19782ae6af94a2918c8fd1f","tag":"autogenerated:zuul:gate","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2025-12-09 08:59:23.000000000","message":"Patch Set 4: -Verified\n\nStarting gate jobs.","accounts_in_message":[],"_revision_number":4},{"id":"a1979c8991295493459f0572c7e6b566eeb0b89b","tag":"autogenerated:zuul:gate","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2025-12-09 10:50:56.000000000","message":"Patch Set 4: Verified+2\n\nBuild succeeded (gate pipeline).\nhttps://zuul.opendev.org/t/openstack/buildset/e270f926db74481d970ccca980be16c6\n\n- openstack-ansible-deploy-hosts_metal-centos-10-stream https://zuul.opendev.org/t/openstack/build/d6b72c186b8347809868ee4de6430bef : SUCCESS in 16m 05s\n- openstack-ansible-deploy-hosts_metal-debian-bookworm https://zuul.opendev.org/t/openstack/build/9d665347af2f41ca8f27f09664dda047 : SUCCESS in 14m 39s\n- openstack-ansible-deploy-hosts_metal-debian-trixie https://zuul.opendev.org/t/openstack/build/49eaf0df51bf4621b840c81e347aa675 : SUCCESS in 12m 52s\n- openstack-ansible-deploy-hosts_metal-rockylinux-10 https://zuul.opendev.org/t/openstack/build/f3bbbdcabc9149bea3e1889a9a4ff336 : SUCCESS in 16m 43s\n- openstack-ansible-deploy-hosts_metal-ubuntu-noble https://zuul.opendev.org/t/openstack/build/26de4bef35474f1fbf70e50a6acf2eac : SUCCESS in 11m 35s\n- openstack-ansible-deploy-hosts_distro_lxc-ubuntu-noble https://zuul.opendev.org/t/openstack/build/1cb4a4f8100b4ca0b64b331c12cdead4 : SUCCESS in 21m 28s\n- openstack-ansible-upgrade-infra_lxc-ubuntu-noble https://zuul.opendev.org/t/openstack/build/26657724910e45e78bbec68ef091b17a : SUCCESS in 1h 46m 58s\n- openstack-tox-docs https://zuul.opendev.org/t/openstack/build/cd200004ff0147e597877f8be6141094 : SUCCESS in 6m 11s\n- openstack-ansible-linters-ubuntu-noble https://zuul.opendev.org/t/openstack/build/33401b3dc41f4b50a51a8d967e58d635 : SUCCESS in 6m 59s","accounts_in_message":[],"_revision_number":4},{"id":"ca471965dee875fba14d7c686b12e3ada53a6b1a","tag":"autogenerated:gerrit:merged","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2025-12-09 10:50:56.000000000","message":"Change has been successfully merged","accounts_in_message":[],"_revision_number":4},{"id":"aecf0050a103bcf4113cbdfb5361a4f20e78327d","tag":"autogenerated:zuul:promote","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2025-12-09 10:51:48.000000000","message":"Patch Set 4:\n\nBuild succeeded (promote pipeline).\nhttps://zuul.opendev.org/t/openstack/buildset/f7b7f4ba28634ce58550fa2eefeaea16\n\n- promote-openstack-tox-docs https://zuul.opendev.org/t/openstack/build/45f64d64994a4e5694d8563b4ad5901a : SUCCESS in 39s","accounts_in_message":[],"_revision_number":4}],"current_revision_number":4,"current_revision":"57dbf72d949f59ee2fa8cc063f69ed96829c3104","revisions":{"241da80a3d6a40700bc03e1491dacd9b0097375b":{"kind":"REWORK","_number":1,"created":"2025-12-08 18:38:15.000000000","uploader":{"_account_id":37598,"name":"Ivan Anfimov","display_name":"Ivan Anfimov","email":"lazekteam@gmail.com","username":"anfimovir"},"ref":"refs/changes/69/970169/1","fetch":{"anonymous http":{"url":"https://review.opendev.org/openstack/openstack-ansible-openstack_hosts","ref":"refs/changes/69/970169/1","commands":{"Checkout":"git fetch https://review.opendev.org/openstack/openstack-ansible-openstack_hosts refs/changes/69/970169/1 \u0026\u0026 git checkout FETCH_HEAD","Cherry Pick":"git fetch https://review.opendev.org/openstack/openstack-ansible-openstack_hosts refs/changes/69/970169/1 \u0026\u0026 git cherry-pick FETCH_HEAD","Format Patch":"git fetch https://review.opendev.org/openstack/openstack-ansible-openstack_hosts refs/changes/69/970169/1 \u0026\u0026 git format-patch -1 --stdout FETCH_HEAD","Pull":"git pull https://review.opendev.org/openstack/openstack-ansible-openstack_hosts refs/changes/69/970169/1"}}},"commit":{"parents":[{"commit":"f220ab8fbed37a338c5191ebbcb7bec4e7c52bc6","subject":"tox: remove linters and functional from envlist","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/openstack-ansible-openstack_hosts/commit/f220ab8fbed37a338c5191ebbcb7bec4e7c52bc6"}]}],"author":{"name":"Ivan Anfimov","email":"lazekteam@gmail.com","date":"2025-12-08 18:38:15.000000000","tz":0},"committer":{"name":"Ivan Anfimov","email":"lazekteam@gmail.com","date":"2025-12-08 18:38:15.000000000","tz":0},"subject":"wip","message":"wip\n\nChange-Id: I0fa3161d67a4cedffa889964261f752b0d65a999\nSigned-off-by: Ivan Anfimov \u003clazekteam@gmail.com\u003e\n","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/openstack-ansible-openstack_hosts/commit/241da80a3d6a40700bc03e1491dacd9b0097375b"}],"resolve_conflicts_web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/openstack-ansible-openstack_hosts/commit/241da80a3d6a40700bc03e1491dacd9b0097375b"}]},"branch":"refs/heads/master"},"3544d4b824ca8a5ba0c29063ceb6c9233e365b05":{"kind":"REWORK","_number":2,"created":"2025-12-08 18:38:43.000000000","uploader":{"_account_id":37598,"name":"Ivan Anfimov","display_name":"Ivan Anfimov","email":"lazekteam@gmail.com","username":"anfimovir"},"ref":"refs/changes/69/970169/2","fetch":{"anonymous http":{"url":"https://review.opendev.org/openstack/openstack-ansible-openstack_hosts","ref":"refs/changes/69/970169/2","commands":{"Checkout":"git fetch https://review.opendev.org/openstack/openstack-ansible-openstack_hosts refs/changes/69/970169/2 \u0026\u0026 git checkout FETCH_HEAD","Cherry Pick":"git fetch https://review.opendev.org/openstack/openstack-ansible-openstack_hosts refs/changes/69/970169/2 \u0026\u0026 git cherry-pick FETCH_HEAD","Format Patch":"git fetch https://review.opendev.org/openstack/openstack-ansible-openstack_hosts refs/changes/69/970169/2 \u0026\u0026 git format-patch -1 --stdout FETCH_HEAD","Pull":"git pull https://review.opendev.org/openstack/openstack-ansible-openstack_hosts refs/changes/69/970169/2"}}},"commit":{"parents":[{"commit":"f220ab8fbed37a338c5191ebbcb7bec4e7c52bc6","subject":"tox: remove linters and functional from envlist","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/openstack-ansible-openstack_hosts/commit/f220ab8fbed37a338c5191ebbcb7bec4e7c52bc6"}]}],"author":{"name":"Ivan Anfimov","email":"lazekteam@gmail.com","date":"2025-12-08 18:38:15.000000000","tz":0},"committer":{"name":"Ivan Anfimov","email":"lazekteam@gmail.com","date":"2025-12-08 18:38:40.000000000","tz":0},"subject":"wip","message":"wip\n\nChange-Id: I0fa3161d67a4cedffa889964261f752b0d65a999\nSigned-off-by: Ivan Anfimov \u003clazekteam@gmail.com\u003e\n","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/openstack-ansible-openstack_hosts/commit/3544d4b824ca8a5ba0c29063ceb6c9233e365b05"}],"resolve_conflicts_web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/openstack-ansible-openstack_hosts/commit/3544d4b824ca8a5ba0c29063ceb6c9233e365b05"}]},"branch":"refs/heads/master"},"eef65a543ab7b76b8d5dfcdfcc02b6ca8a56b3ef":{"kind":"NO_CODE_CHANGE","_number":3,"created":"2025-12-08 19:42:56.000000000","uploader":{"_account_id":37598,"name":"Ivan Anfimov","display_name":"Ivan Anfimov","email":"lazekteam@gmail.com","username":"anfimovir"},"ref":"refs/changes/69/970169/3","fetch":{"anonymous http":{"url":"https://review.opendev.org/openstack/openstack-ansible-openstack_hosts","ref":"refs/changes/69/970169/3","commands":{"Checkout":"git fetch https://review.opendev.org/openstack/openstack-ansible-openstack_hosts refs/changes/69/970169/3 \u0026\u0026 git checkout FETCH_HEAD","Cherry Pick":"git fetch https://review.opendev.org/openstack/openstack-ansible-openstack_hosts refs/changes/69/970169/3 \u0026\u0026 git cherry-pick FETCH_HEAD","Format Patch":"git fetch https://review.opendev.org/openstack/openstack-ansible-openstack_hosts refs/changes/69/970169/3 \u0026\u0026 git format-patch -1 --stdout FETCH_HEAD","Pull":"git pull https://review.opendev.org/openstack/openstack-ansible-openstack_hosts refs/changes/69/970169/3"}}},"commit":{"parents":[{"commit":"f220ab8fbed37a338c5191ebbcb7bec4e7c52bc6","subject":"tox: remove linters and functional from envlist","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/openstack-ansible-openstack_hosts/commit/f220ab8fbed37a338c5191ebbcb7bec4e7c52bc6"}]}],"author":{"name":"Ivan Anfimov","email":"lazekteam@gmail.com","date":"2025-12-08 18:38:15.000000000","tz":0},"committer":{"name":"Ivan Anfimov","email":"lazekteam@gmail.com","date":"2025-12-08 19:42:56.000000000","tz":0},"subject":"Remove outdated configure or yum","message":"Remove outdated configure or yum\n\nWe now use dnf in RHEL10.\n\nChange-Id: I0fa3161d67a4cedffa889964261f752b0d65a999\nSigned-off-by: Ivan Anfimov \u003clazekteam@gmail.com\u003e\n","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/openstack-ansible-openstack_hosts/commit/eef65a543ab7b76b8d5dfcdfcc02b6ca8a56b3ef"}],"resolve_conflicts_web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/openstack-ansible-openstack_hosts/commit/eef65a543ab7b76b8d5dfcdfcc02b6ca8a56b3ef"}]},"branch":"refs/heads/master","description":"Edit commit message"},"57dbf72d949f59ee2fa8cc063f69ed96829c3104":{"kind":"NO_CODE_CHANGE","_number":4,"created":"2025-12-08 19:44:42.000000000","uploader":{"_account_id":37598,"name":"Ivan Anfimov","display_name":"Ivan Anfimov","email":"lazekteam@gmail.com","username":"anfimovir"},"ref":"refs/changes/69/970169/4","fetch":{"anonymous http":{"url":"https://review.opendev.org/openstack/openstack-ansible-openstack_hosts","ref":"refs/changes/69/970169/4","commands":{"Checkout":"git fetch https://review.opendev.org/openstack/openstack-ansible-openstack_hosts refs/changes/69/970169/4 \u0026\u0026 git checkout FETCH_HEAD","Cherry Pick":"git fetch https://review.opendev.org/openstack/openstack-ansible-openstack_hosts refs/changes/69/970169/4 \u0026\u0026 git cherry-pick FETCH_HEAD","Format Patch":"git fetch https://review.opendev.org/openstack/openstack-ansible-openstack_hosts refs/changes/69/970169/4 \u0026\u0026 git format-patch -1 --stdout FETCH_HEAD","Pull":"git pull https://review.opendev.org/openstack/openstack-ansible-openstack_hosts refs/changes/69/970169/4"}}},"commit":{"parents":[{"commit":"f220ab8fbed37a338c5191ebbcb7bec4e7c52bc6","subject":"tox: remove linters and functional from envlist","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/openstack-ansible-openstack_hosts/commit/f220ab8fbed37a338c5191ebbcb7bec4e7c52bc6"}]}],"author":{"name":"Ivan Anfimov","email":"lazekteam@gmail.com","date":"2025-12-08 18:38:15.000000000","tz":0},"committer":{"name":"Ivan Anfimov","email":"lazekteam@gmail.com","date":"2025-12-08 19:44:42.000000000","tz":0},"subject":"Remove outdated configure for yum","message":"Remove outdated configure for yum\n\nWe now use dnf in RHEL10.\n\nChange-Id: I0fa3161d67a4cedffa889964261f752b0d65a999\nSigned-off-by: Ivan Anfimov \u003clazekteam@gmail.com\u003e\n","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/openstack-ansible-openstack_hosts/commit/57dbf72d949f59ee2fa8cc063f69ed96829c3104"}],"resolve_conflicts_web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/openstack-ansible-openstack_hosts/commit/57dbf72d949f59ee2fa8cc063f69ed96829c3104"}]},"branch":"refs/heads/master","description":"Edit commit message"}},"requirements":[],"submit_records":[{"rule_name":"gerrit~DefaultSubmitRule","status":"CLOSED","labels":[{"label":"Verified","status":"MAY","applied_by":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]}},{"label":"Code-Review","status":"MAY","applied_by":{"_account_id":31542,"name":"Andrew Bonney","email":"andrew.bonney@bbc.co.uk","username":"andrewbonney"}},{"label":"Workflow","status":"MAY","applied_by":{"_account_id":31542,"name":"Andrew Bonney","email":"andrew.bonney@bbc.co.uk","username":"andrewbonney"}},{"label":"Backport-Candidate","status":"MAY"}]}],"submit_requirements":[{"name":"Verified","description":"Verified in gate by CI","status":"SATISFIED","is_legacy":false,"submittability_expression_result":{"expression":"label:Verified\u003dMAX AND -label:Verified\u003dMIN","fulfilled":true,"status":"PASS","passing_atoms":["label:Verified\u003dMAX"],"failing_atoms":["label:Verified\u003dMIN"],"atom_explanations":{}}},{"name":"Backport-Candidate","description":"Backport candidate status","status":"NOT_APPLICABLE","is_legacy":false,"applicability_expression_result":{"fulfilled":false,"status":"FAIL"},"submittability_expression_result":{"expression":"is:true","fulfilled":true,"status":"NOT_EVALUATED","passing_atoms":[],"failing_atoms":[],"atom_explanations":{}}},{"name":"Code-Review","description":"Code reviewed by core reviewer","status":"SATISFIED","is_legacy":false,"submittability_expression_result":{"expression":"label:Code-Review\u003dMAX AND -label:Code-Review\u003dMIN","fulfilled":true,"status":"PASS","passing_atoms":["label:Code-Review\u003dMAX"],"failing_atoms":["label:Code-Review\u003dMIN"],"atom_explanations":{}}},{"name":"Workflow","description":"Approved for gate by core reviewer","status":"SATISFIED","is_legacy":false,"submittability_expression_result":{"expression":"label:Workflow\u003dMAX AND -label:Workflow\u003dMIN","fulfilled":true,"status":"PASS","passing_atoms":["label:Workflow\u003dMAX"],"failing_atoms":["label:Workflow\u003dMIN"],"atom_explanations":{}}}]}
