)]}'
{"id":"openstack%2Fopenstack-ansible-ops~611741","triplet_id":"openstack%2Fopenstack-ansible-ops~master~I75de7063164959041193526e519682ba80185250","project":"openstack/openstack-ansible-ops","branch":"master","hashtags":[],"change_id":"I75de7063164959041193526e519682ba80185250","subject":"Additional cleanup and simplification","status":"MERGED","created":"2018-10-19 02:17:53.000000000","updated":"2018-10-19 04:47:56.000000000","submitted":"2018-10-19 04:47:56.000000000","submitter":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"total_comment_count":0,"unresolved_comment_count":0,"has_review_started":true,"submission_id":"611741-1539924476008-3eb7f163","meta_rev_id":"e435fbd377a5d3f3945b8f09914bac500679fe16","_number":611741,"virtual_id_number":611741,"owner":{"_account_id":7353,"name":"Kevin Carter","email":"kevin@cloudnull.com","username":"cloudnull"},"actions":{},"labels":{"Verified":{"approved":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"all":[{"value":0,"_account_id":7353,"name":"Kevin Carter","email":"kevin@cloudnull.com","username":"cloudnull"},{"value":2,"date":"2018-10-19 04:47:55.000000000","permitted_voting_range":{"min":2,"max":2},"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]}],"values":{"-2":"Fails","-1":"Doesn\u0027t seem to work"," 0":"No score","+1":"Works for me","+2":"Verified"},"description":"","default_value":0,"optional":true},"Code-Review":{"approved":{"_account_id":7353,"name":"Kevin Carter","email":"kevin@cloudnull.com","username":"cloudnull"},"all":[{"value":2,"date":"2018-10-19 04:37:33.000000000","permitted_voting_range":{"min":2,"max":2},"_account_id":7353,"name":"Kevin Carter","email":"kevin@cloudnull.com","username":"cloudnull"},{"value":0,"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]}],"values":{"-2":"Do not merge","-1":"This patch needs further work before it can be merged"," 0":"No score","+1":"Looks good to me, but someone else must approve","+2":"Looks good to me (core reviewer)"},"description":"","default_value":0,"optional":true},"Workflow":{"approved":{"_account_id":7353,"name":"Kevin Carter","email":"kevin@cloudnull.com","username":"cloudnull"},"all":[{"value":1,"date":"2018-10-19 04:37:33.000000000","permitted_voting_range":{"min":1,"max":1},"_account_id":7353,"name":"Kevin Carter","email":"kevin@cloudnull.com","username":"cloudnull"},{"value":0,"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]}],"values":{"-1":"Work in progress"," 0":"Ready for reviews","+1":"Approved"},"description":"","default_value":0,"optional":true},"Backport-Candidate":{"all":[{"value":0,"_account_id":7353,"name":"Kevin Carter","email":"kevin@cloudnull.com","username":"cloudnull"},{"value":0,"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]}],"values":{"-1":"Not A Backport Candidate"," 0":"Backport Review Needed","+1":"Proposed Backport"},"description":"","default_value":0,"optional":true}},"removable_reviewers":[],"reviewers":{"REVIEWER":[{"_account_id":7353,"name":"Kevin Carter","email":"kevin@cloudnull.com","username":"cloudnull"},{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]}]},"pending_reviewers":{},"reviewer_updates":[{"updated":"2018-10-19 04:47:55.000000000","updated_by":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"reviewer":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"state":"REVIEWER"}],"messages":[{"id":"a8021a36ca9fa16056a25bbbba06e538ff0008ff","author":{"_account_id":7353,"name":"Kevin Carter","email":"kevin@cloudnull.com","username":"cloudnull"},"date":"2018-10-19 02:17:53.000000000","message":"Uploaded patch set 1.","accounts_in_message":[],"_revision_number":1},{"id":"e1a8103041a0d71e379c78ea3341434af0119b9c","author":{"_account_id":7353,"name":"Kevin Carter","email":"kevin@cloudnull.com","username":"cloudnull"},"date":"2018-10-19 02:29:15.000000000","message":"Uploaded patch set 2.","accounts_in_message":[],"_revision_number":2},{"id":"773283262c55e747e1e16ad59468a5883af0cb85","author":{"_account_id":7353,"name":"Kevin Carter","email":"kevin@cloudnull.com","username":"cloudnull"},"date":"2018-10-19 02:30:36.000000000","message":"Uploaded patch set 3.","accounts_in_message":[],"_revision_number":3},{"id":"7ddc66d928636e96fdc4beda60ce433b81514031","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2018-10-19 03:22:21.000000000","message":"Patch Set 3: Verified-1\n\nBuild failed (check pipeline).  For information on how to proceed, see\nhttp://docs.openstack.org/infra/manual/developers.html#automated-testing\n\n\n- openstack-ansible-linters http://logs.openstack.org/41/611741/3/check/openstack-ansible-linters/21c5825/ : SUCCESS in 6m 26s\n- openstack-ansible-functional-centos-7 http://logs.openstack.org/41/611741/3/check/openstack-ansible-functional-centos-7/256e268/ : SUCCESS in 3m 33s\n- openstack-ansible-functional-opensuse-150 http://logs.openstack.org/41/611741/3/check/openstack-ansible-functional-opensuse-150/282f1a5/ : SUCCESS in 4m 06s (non-voting)\n- openstack-ansible-functional-opensuse-423 http://logs.openstack.org/41/611741/3/check/openstack-ansible-functional-opensuse-423/4427052/ : SUCCESS in 4m 48s\n- openstack-ansible-functional-ubuntu-bionic http://logs.openstack.org/41/611741/3/check/openstack-ansible-functional-ubuntu-bionic/4dade2e/ : SUCCESS in 3m 49s\n- openstack-tox-docs http://logs.openstack.org/41/611741/3/check/openstack-tox-docs/f67636a/html/ : SUCCESS in 2m 38s\n- openstack-ansible-ops:osquery-ubuntu-xenial http://logs.openstack.org/41/611741/3/check/openstack-ansible-ops:osquery-ubuntu-xenial/c202fcf/ : FAILURE in 7m 00s\n- openstack-ansible-ops:osquery-ubuntu-xenial-clustered http://logs.openstack.org/41/611741/3/check/openstack-ansible-ops:osquery-ubuntu-xenial-clustered/e689b3a/ : TIMED_OUT in 50m 27s\n- openstack-ansible-ops:osquery-ubuntu-bionic http://logs.openstack.org/41/611741/3/check/openstack-ansible-ops:osquery-ubuntu-bionic/828f8e7/ : FAILURE in 6m 42s\n- openstack-ansible-ops:osquery-ubuntu-bionic-clustered http://logs.openstack.org/41/611741/3/check/openstack-ansible-ops:osquery-ubuntu-bionic-clustered/26d4403/ : TIMED_OUT in 30m 41s","accounts_in_message":[],"_revision_number":3},{"id":"ac7394428c6428418994143ef7194316fd1b1943","author":{"_account_id":7353,"name":"Kevin Carter","email":"kevin@cloudnull.com","username":"cloudnull"},"date":"2018-10-19 03:53:52.000000000","message":"Uploaded patch set 4.","accounts_in_message":[],"_revision_number":4},{"id":"556c36d51c80bd963545697270a94c90c795a8c1","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2018-10-19 04:13:16.000000000","message":"Patch Set 4: Verified+1\n\nBuild succeeded (check pipeline).\n\n- openstack-ansible-linters http://logs.openstack.org/41/611741/4/check/openstack-ansible-linters/8762543/ : SUCCESS in 6m 05s\n- openstack-ansible-functional-centos-7 http://logs.openstack.org/41/611741/4/check/openstack-ansible-functional-centos-7/e35fa3c/ : SUCCESS in 5m 25s\n- openstack-ansible-functional-opensuse-150 http://logs.openstack.org/41/611741/4/check/openstack-ansible-functional-opensuse-150/e506d3c/ : SUCCESS in 5m 49s (non-voting)\n- openstack-ansible-functional-opensuse-423 http://logs.openstack.org/41/611741/4/check/openstack-ansible-functional-opensuse-423/29e7502/ : SUCCESS in 5m 45s\n- openstack-ansible-functional-ubuntu-bionic http://logs.openstack.org/41/611741/4/check/openstack-ansible-functional-ubuntu-bionic/01580d2/ : SUCCESS in 3m 18s\n- openstack-tox-docs http://logs.openstack.org/41/611741/4/check/openstack-tox-docs/7eedee4/html/ : SUCCESS in 3m 30s\n- openstack-ansible-ops:osquery-ubuntu-xenial http://logs.openstack.org/41/611741/4/check/openstack-ansible-ops:osquery-ubuntu-xenial/ad6085c/ : SUCCESS in 5m 34s\n- openstack-ansible-ops:osquery-ubuntu-xenial-clustered http://logs.openstack.org/41/611741/4/check/openstack-ansible-ops:osquery-ubuntu-xenial-clustered/5cb7a83/ : SUCCESS in 12m 46s\n- openstack-ansible-ops:osquery-ubuntu-bionic http://logs.openstack.org/41/611741/4/check/openstack-ansible-ops:osquery-ubuntu-bionic/f4425c5/ : SUCCESS in 5m 18s\n- openstack-ansible-ops:osquery-ubuntu-bionic-clustered http://logs.openstack.org/41/611741/4/check/openstack-ansible-ops:osquery-ubuntu-bionic-clustered/e54943a/ : SUCCESS in 14m 04s","accounts_in_message":[],"_revision_number":4},{"id":"bcb7a97b531d195cfab4d6972080dc6d52b271ab","author":{"_account_id":7353,"name":"Kevin Carter","email":"kevin@cloudnull.com","username":"cloudnull"},"date":"2018-10-19 04:37:33.000000000","message":"Patch Set 4: Code-Review+2 Workflow+1","accounts_in_message":[],"_revision_number":4},{"id":"eb09c61bbc06c892ba925ec0e167abafa8eba1a6","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2018-10-19 04:37:44.000000000","message":"Patch Set 4: -Verified\n\nStarting gate jobs.","accounts_in_message":[],"_revision_number":4},{"id":"e317beda6ed7393a3647e8c52f31a02595eeb875","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2018-10-19 04:47:55.000000000","message":"Patch Set 4: Verified+2\n\nBuild succeeded (gate pipeline).\n\n- openstack-ansible-linters http://logs.openstack.org/41/611741/4/gate/openstack-ansible-linters/ae39bd8/ : SUCCESS in 5m 45s\n- openstack-ansible-functional-centos-7 http://logs.openstack.org/41/611741/4/gate/openstack-ansible-functional-centos-7/bab5344/ : SUCCESS in 4m 09s\n- openstack-ansible-functional-opensuse-423 http://logs.openstack.org/41/611741/4/gate/openstack-ansible-functional-opensuse-423/9b92e9c/ : SUCCESS in 4m 04s\n- openstack-ansible-functional-ubuntu-bionic http://logs.openstack.org/41/611741/4/gate/openstack-ansible-functional-ubuntu-bionic/ac772d3/ : SUCCESS in 3m 52s\n- openstack-tox-docs http://logs.openstack.org/41/611741/4/gate/openstack-tox-docs/71b0dcc/html/ : SUCCESS in 2m 40s\n- openstack-ansible-ops:osquery-ubuntu-bionic http://logs.openstack.org/41/611741/4/gate/openstack-ansible-ops:osquery-ubuntu-bionic/5afdc54/ : SUCCESS in 6m 35s","accounts_in_message":[],"_revision_number":4},{"id":"e435fbd377a5d3f3945b8f09914bac500679fe16","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2018-10-19 04:47:56.000000000","message":"Change has been successfully merged by Zuul","accounts_in_message":[],"_revision_number":4}],"current_revision_number":4,"current_revision":"fd2005baf4e9efa0a9996e051e1b7fbf8d3a6cf9","revisions":{"221409b0430fb41c7d4fb884363c03cafed9fa2f":{"kind":"REWORK","_number":1,"created":"2018-10-19 02:17:53.000000000","uploader":{"_account_id":7353,"name":"Kevin Carter","email":"kevin@cloudnull.com","username":"cloudnull"},"ref":"refs/changes/41/611741/1","fetch":{"anonymous http":{"url":"https://review.opendev.org/openstack/openstack-ansible-ops","ref":"refs/changes/41/611741/1","commands":{"Checkout":"git fetch https://review.opendev.org/openstack/openstack-ansible-ops refs/changes/41/611741/1 \u0026\u0026 git checkout FETCH_HEAD","Cherry Pick":"git fetch https://review.opendev.org/openstack/openstack-ansible-ops refs/changes/41/611741/1 \u0026\u0026 git cherry-pick FETCH_HEAD","Format Patch":"git fetch https://review.opendev.org/openstack/openstack-ansible-ops refs/changes/41/611741/1 \u0026\u0026 git format-patch -1 --stdout FETCH_HEAD","Pull":"git pull https://review.opendev.org/openstack/openstack-ansible-ops refs/changes/41/611741/1"}}},"commit":{"parents":[{"commit":"b65bd6178f004818c11b0b7c13b406d0841ff99f","subject":"Merge \"Cleanup the osquery role\"","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/openstack-ansible-ops/commit/b65bd6178f004818c11b0b7c13b406d0841ff99f"}]}],"author":{"name":"Kevin Carter","email":"kevin.carter@rackspace.com","date":"2018-10-19 02:13:51.000000000","tz":-300},"committer":{"name":"Kevin Carter","email":"kevin.carter@rackspace.com","date":"2018-10-19 02:17:44.000000000","tz":-300},"subject":"Additional cleanup and simplification","message":"Additional cleanup and simplification\n\nThe role is further being pruned and made more efficient.\n\nChange-Id: I75de7063164959041193526e519682ba80185250\nSigned-off-by: Kevin Carter \u003ckevin.carter@rackspace.com\u003e\n","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/openstack-ansible-ops/commit/221409b0430fb41c7d4fb884363c03cafed9fa2f"}],"resolve_conflicts_web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/openstack-ansible-ops/commit/221409b0430fb41c7d4fb884363c03cafed9fa2f"}]},"parents_data":[{"branch_name":"refs/heads/master","commit_id":"b65bd6178f004818c11b0b7c13b406d0841ff99f","is_merged_in_target_branch":true}],"branch":"refs/heads/master"},"c16edae79414d20f1c7b982e5918910425fe8c39":{"kind":"REWORK","_number":2,"created":"2018-10-19 02:29:15.000000000","uploader":{"_account_id":7353,"name":"Kevin Carter","email":"kevin@cloudnull.com","username":"cloudnull"},"ref":"refs/changes/41/611741/2","fetch":{"anonymous http":{"url":"https://review.opendev.org/openstack/openstack-ansible-ops","ref":"refs/changes/41/611741/2","commands":{"Checkout":"git fetch https://review.opendev.org/openstack/openstack-ansible-ops refs/changes/41/611741/2 \u0026\u0026 git checkout FETCH_HEAD","Cherry Pick":"git fetch https://review.opendev.org/openstack/openstack-ansible-ops refs/changes/41/611741/2 \u0026\u0026 git cherry-pick FETCH_HEAD","Format Patch":"git fetch https://review.opendev.org/openstack/openstack-ansible-ops refs/changes/41/611741/2 \u0026\u0026 git format-patch -1 --stdout FETCH_HEAD","Pull":"git pull https://review.opendev.org/openstack/openstack-ansible-ops refs/changes/41/611741/2"}}},"commit":{"parents":[{"commit":"e0e85797852a59ba74d7d94c17f966ec6c51282d","subject":"Additional playbook cleanup and use stable release","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/openstack-ansible-ops/commit/e0e85797852a59ba74d7d94c17f966ec6c51282d"}]}],"author":{"name":"Kevin Carter","email":"kevin.carter@rackspace.com","date":"2018-10-19 02:13:51.000000000","tz":-300},"committer":{"name":"Kevin Carter","email":"kevin.carter@rackspace.com","date":"2018-10-19 02:29:05.000000000","tz":-300},"subject":"Additional cleanup and simplification","message":"Additional cleanup and simplification\n\nThe role is further being pruned and made more efficient.\n\nChange-Id: I75de7063164959041193526e519682ba80185250\nSigned-off-by: Kevin Carter \u003ckevin.carter@rackspace.com\u003e\n","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/openstack-ansible-ops/commit/c16edae79414d20f1c7b982e5918910425fe8c39"}],"resolve_conflicts_web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/openstack-ansible-ops/commit/c16edae79414d20f1c7b982e5918910425fe8c39"}]},"parents_data":[{"branch_name":"refs/heads/master","commit_id":"e0e85797852a59ba74d7d94c17f966ec6c51282d","is_merged_in_target_branch":true,"change_id":"Ie488fe42e98bd96f5777c1c893fc79e50a2eddee","change_number":611661,"patch_set_number":4,"change_status":"MERGED"}],"branch":"refs/heads/master"},"3ba56d33f527b808d7a3f58d82f4bebf87ff734d":{"kind":"REWORK","_number":3,"created":"2018-10-19 02:30:36.000000000","uploader":{"_account_id":7353,"name":"Kevin Carter","email":"kevin@cloudnull.com","username":"cloudnull"},"ref":"refs/changes/41/611741/3","fetch":{"anonymous http":{"url":"https://review.opendev.org/openstack/openstack-ansible-ops","ref":"refs/changes/41/611741/3","commands":{"Checkout":"git fetch https://review.opendev.org/openstack/openstack-ansible-ops refs/changes/41/611741/3 \u0026\u0026 git checkout FETCH_HEAD","Cherry Pick":"git fetch https://review.opendev.org/openstack/openstack-ansible-ops refs/changes/41/611741/3 \u0026\u0026 git cherry-pick FETCH_HEAD","Format Patch":"git fetch https://review.opendev.org/openstack/openstack-ansible-ops refs/changes/41/611741/3 \u0026\u0026 git format-patch -1 --stdout FETCH_HEAD","Pull":"git pull https://review.opendev.org/openstack/openstack-ansible-ops refs/changes/41/611741/3"}}},"commit":{"parents":[{"commit":"e0e85797852a59ba74d7d94c17f966ec6c51282d","subject":"Additional playbook cleanup and use stable release","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/openstack-ansible-ops/commit/e0e85797852a59ba74d7d94c17f966ec6c51282d"}]}],"author":{"name":"Kevin Carter","email":"kevin.carter@rackspace.com","date":"2018-10-19 02:13:51.000000000","tz":-300},"committer":{"name":"Kevin Carter","email":"kevin.carter@rackspace.com","date":"2018-10-19 02:30:29.000000000","tz":-300},"subject":"Additional cleanup and simplification","message":"Additional cleanup and simplification\n\nThe role is further being pruned and made more efficient.\n\nChange-Id: I75de7063164959041193526e519682ba80185250\nSigned-off-by: Kevin Carter \u003ckevin.carter@rackspace.com\u003e\n","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/openstack-ansible-ops/commit/3ba56d33f527b808d7a3f58d82f4bebf87ff734d"}],"resolve_conflicts_web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/openstack-ansible-ops/commit/3ba56d33f527b808d7a3f58d82f4bebf87ff734d"}]},"parents_data":[{"branch_name":"refs/heads/master","commit_id":"e0e85797852a59ba74d7d94c17f966ec6c51282d","is_merged_in_target_branch":true,"change_id":"Ie488fe42e98bd96f5777c1c893fc79e50a2eddee","change_number":611661,"patch_set_number":4,"change_status":"MERGED"}],"branch":"refs/heads/master"},"fd2005baf4e9efa0a9996e051e1b7fbf8d3a6cf9":{"kind":"REWORK","_number":4,"created":"2018-10-19 03:53:52.000000000","uploader":{"_account_id":7353,"name":"Kevin Carter","email":"kevin@cloudnull.com","username":"cloudnull"},"ref":"refs/changes/41/611741/4","fetch":{"anonymous http":{"url":"https://review.opendev.org/openstack/openstack-ansible-ops","ref":"refs/changes/41/611741/4","commands":{"Checkout":"git fetch https://review.opendev.org/openstack/openstack-ansible-ops refs/changes/41/611741/4 \u0026\u0026 git checkout FETCH_HEAD","Cherry Pick":"git fetch https://review.opendev.org/openstack/openstack-ansible-ops refs/changes/41/611741/4 \u0026\u0026 git cherry-pick FETCH_HEAD","Format Patch":"git fetch https://review.opendev.org/openstack/openstack-ansible-ops refs/changes/41/611741/4 \u0026\u0026 git format-patch -1 --stdout FETCH_HEAD","Pull":"git pull https://review.opendev.org/openstack/openstack-ansible-ops refs/changes/41/611741/4"}}},"commit":{"parents":[{"commit":"e0e85797852a59ba74d7d94c17f966ec6c51282d","subject":"Additional playbook cleanup and use stable release","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/openstack-ansible-ops/commit/e0e85797852a59ba74d7d94c17f966ec6c51282d"}]}],"author":{"name":"Kevin Carter","email":"kevin.carter@rackspace.com","date":"2018-10-19 02:13:51.000000000","tz":-300},"committer":{"name":"Kevin Carter","email":"kevin.carter@rackspace.com","date":"2018-10-19 03:51:04.000000000","tz":-300},"subject":"Additional cleanup and simplification","message":"Additional cleanup and simplification\n\nThe role is further being pruned and made more efficient.\n\nChange-Id: I75de7063164959041193526e519682ba80185250\nSigned-off-by: Kevin Carter \u003ckevin.carter@rackspace.com\u003e\n","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/openstack-ansible-ops/commit/fd2005baf4e9efa0a9996e051e1b7fbf8d3a6cf9"}],"resolve_conflicts_web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/openstack-ansible-ops/commit/fd2005baf4e9efa0a9996e051e1b7fbf8d3a6cf9"}]},"parents_data":[{"branch_name":"refs/heads/master","commit_id":"e0e85797852a59ba74d7d94c17f966ec6c51282d","is_merged_in_target_branch":true,"change_id":"Ie488fe42e98bd96f5777c1c893fc79e50a2eddee","change_number":611661,"patch_set_number":4,"change_status":"MERGED"}],"branch":"refs/heads/master"}},"requirements":[],"submit_records":[],"submit_requirements":[]}
