)]}'
{"id":"openstack%2Fopenstack-ansible-os_adjutant~835697","triplet_id":"openstack%2Fopenstack-ansible-os_adjutant~master~Ifedfd0e702c91820337c0e61b7d780bab18f4892","project":"openstack/openstack-ansible-os_adjutant","branch":"master","topic":"openstack/openstack-ansible-tests/sync-tests","attention_set":{},"removed_from_attention_set":{"25023":{"account":{"_account_id":25023,"name":"Jonathan Rosser","email":"jonathan.rosser@rd.bbc.co.uk","username":"jrosser"},"last_update":"2022-04-04 16:59:54.000000000","reason":"Change was submitted"}},"hashtags":[],"change_id":"Ifedfd0e702c91820337c0e61b7d780bab18f4892","subject":"Updated from OpenStack Ansible Tests","status":"MERGED","created":"2022-03-29 17:17:45.000000000","updated":"2022-04-04 17:01:18.000000000","submitted":"2022-04-04 16:59:54.000000000","submitter":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"total_comment_count":0,"unresolved_comment_count":0,"has_review_started":true,"submission_id":"835697-openstack/openstack-ansible-tests/sync-tests","meta_rev_id":"e3408e97169d1fa640a8c170fb9bff90a4226600","_number":835697,"virtual_id_number":835697,"owner":{"_account_id":11131,"name":"OpenStack Proposal Bot","username":"proposal-bot","tags":["SERVICE_USER"]},"actions":{},"labels":{"Verified":{"approved":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"all":[{"value":0,"_account_id":31542,"name":"Andrew Bonney","email":"andrew.bonney@bbc.co.uk","username":"andrewbonney"},{"value":0,"_account_id":28619,"name":"Dmitriy Rabotyagov","email":"noonedeadpunk@gmail.com","username":"noonedeadpunk"},{"tag":"autogenerated:zuul:gate","value":2,"date":"2022-04-04 16:59:54.000000000","permitted_voting_range":{"min":2,"max":2},"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]}],"values":{"-2":"Fails","-1":"Doesn\u0027t seem to work"," 0":"No score","+1":"Works for me","+2":"Verified"},"description":"","default_value":0,"optional":true},"Code-Review":{"approved":{"_account_id":31542,"name":"Andrew Bonney","email":"andrew.bonney@bbc.co.uk","username":"andrewbonney"},"all":[{"value":2,"date":"2022-04-04 15:07:21.000000000","permitted_voting_range":{"min":2,"max":2},"_account_id":31542,"name":"Andrew Bonney","email":"andrew.bonney@bbc.co.uk","username":"andrewbonney"},{"value":2,"date":"2022-04-04 09:32:30.000000000","permitted_voting_range":{"min":2,"max":2},"_account_id":28619,"name":"Dmitriy Rabotyagov","email":"noonedeadpunk@gmail.com","username":"noonedeadpunk"},{"value":0,"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]}],"values":{"-2":"Do not merge","-1":"This patch needs further work before it can be merged"," 0":"No score","+1":"Looks good to me, but someone else must approve","+2":"Looks good to me (core reviewer)"},"description":"","default_value":0,"optional":true},"Workflow":{"approved":{"_account_id":31542,"name":"Andrew Bonney","email":"andrew.bonney@bbc.co.uk","username":"andrewbonney"},"all":[{"value":1,"date":"2022-04-04 15:07:21.000000000","permitted_voting_range":{"min":1,"max":1},"_account_id":31542,"name":"Andrew Bonney","email":"andrew.bonney@bbc.co.uk","username":"andrewbonney"},{"value":0,"_account_id":28619,"name":"Dmitriy Rabotyagov","email":"noonedeadpunk@gmail.com","username":"noonedeadpunk"},{"value":0,"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]}],"values":{"-1":"Work in progress"," 0":"Ready for reviews","+1":"Approved"},"description":"","default_value":0,"optional":true},"Backport-Candidate":{"all":[{"value":0,"_account_id":31542,"name":"Andrew Bonney","email":"andrew.bonney@bbc.co.uk","username":"andrewbonney"},{"value":0,"_account_id":28619,"name":"Dmitriy Rabotyagov","email":"noonedeadpunk@gmail.com","username":"noonedeadpunk"},{"value":0,"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]}],"values":{"-1":"Not A Backport Candidate"," 0":"Backport Review Needed","+1":"Proposed Backport"},"description":"","default_value":0,"optional":true}},"removable_reviewers":[],"reviewers":{"REVIEWER":[{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},{"_account_id":28619,"name":"Dmitriy Rabotyagov","email":"noonedeadpunk@gmail.com","username":"noonedeadpunk"},{"_account_id":31542,"name":"Andrew Bonney","email":"andrew.bonney@bbc.co.uk","username":"andrewbonney"}]},"pending_reviewers":{},"reviewer_updates":[{"updated":"2022-03-29 21:53:26.000000000","updated_by":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"reviewer":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"state":"REVIEWER"},{"updated":"2022-04-04 09:32:30.000000000","updated_by":{"_account_id":28619,"name":"Dmitriy Rabotyagov","email":"noonedeadpunk@gmail.com","username":"noonedeadpunk"},"reviewer":{"_account_id":28619,"name":"Dmitriy Rabotyagov","email":"noonedeadpunk@gmail.com","username":"noonedeadpunk"},"state":"REVIEWER"},{"updated":"2022-04-04 15:07:21.000000000","updated_by":{"_account_id":31542,"name":"Andrew Bonney","email":"andrew.bonney@bbc.co.uk","username":"andrewbonney"},"reviewer":{"_account_id":31542,"name":"Andrew Bonney","email":"andrew.bonney@bbc.co.uk","username":"andrewbonney"},"state":"REVIEWER"}],"messages":[{"id":"c5cb68e7f6964f9d672aad4f83656802a28f62a0","tag":"autogenerated:gerrit:newPatchSet","author":{"_account_id":11131,"name":"OpenStack Proposal Bot","username":"proposal-bot","tags":["SERVICE_USER"]},"date":"2022-03-29 17:17:45.000000000","message":"Uploaded patch set 1.","accounts_in_message":[],"_revision_number":1},{"id":"a8ad59ec957ac47d14d9c8f5fef694ca4a6276c6","tag":"autogenerated:zuul:check","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2022-03-29 21:53:26.000000000","message":"Patch Set 1: Verified-1\n\nBuild failed (check pipeline).  For information on how to proceed, see\nhttps://docs.opendev.org/opendev/infra-manual/latest/developers.html#automated-testing\n\n\n- openstack-ansible-linters-ubuntu-focal https://zuul.opendev.org/t/openstack/build/5e4f4f323d044bfb84318e18d0e301f1 : SUCCESS in 15m 57s\n- openstack-ansible-deploy-aio_metal-debian-bullseye https://zuul.opendev.org/t/openstack/build/d94e0084f1d44369b94f043d6d733977 : SUCCESS in 1h 20m 53s\n- openstack-ansible-deploy-aio_metal-centos-8-stream https://zuul.opendev.org/t/openstack/build/797b9e0d8c9d4251b79a4f521176653a : SUCCESS in 1h 28m 42s\n- openstack-ansible-deploy-aio_metal-ubuntu-focal https://zuul.opendev.org/t/openstack/build/4230f57536824c71a9d2cb9290237de1 : SUCCESS in 1h 35m 02s\n- openstack-ansible-deploy-aio_metal-rockylinux-8 https://zuul.opendev.org/t/openstack/build/3d58eb4ba29f450e9a3aeeb99fc2b298 : SUCCESS in 1h 25m 31s\n- openstack-ansible-upgrade-aio_metal-centos-8-stream https://zuul.opendev.org/t/openstack/build/a7022edaa62c4a38815bfeccbffa6131 : FAILURE in 1h 23m 14s (non-voting)\n- openstack-ansible-upgrade-aio_metal-ubuntu-focal https://zuul.opendev.org/t/openstack/build/e152de3ba87a47b2980458c09949033e : SUCCESS in 2h 19m 16s\n- openstack-tox-docs https://zuul.opendev.org/t/openstack/build/440735911a39409aaabbb774cd170d02 : FAILURE in 6m 24s","accounts_in_message":[],"_revision_number":1},{"id":"a2affad854868bf07457697e5259bc6e92926c6b","tag":"autogenerated:gerrit:newPatchSet","author":{"_account_id":25023,"name":"Jonathan Rosser","email":"jonathan.rosser@rd.bbc.co.uk","username":"jrosser"},"date":"2022-04-04 09:09:52.000000000","message":"Patch Set 2: Patch Set 1 was rebased","accounts_in_message":[],"_revision_number":2},{"id":"a766b3d80363ecdcf17e96e13290e0816da380c4","author":{"_account_id":28619,"name":"Dmitriy Rabotyagov","email":"noonedeadpunk@gmail.com","username":"noonedeadpunk"},"date":"2022-04-04 09:32:30.000000000","message":"Patch Set 2: Code-Review+2","accounts_in_message":[],"_revision_number":2},{"id":"8ca52e850dd914b0dd578079b64b1b3d3232e09d","tag":"autogenerated:zuul:check","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2022-04-04 11:19:07.000000000","message":"Patch Set 2: Verified+1\n\nBuild succeeded (check pipeline).\n\n- openstack-ansible-linters-ubuntu-focal https://zuul.opendev.org/t/openstack/build/4147ee3c67b546ee88173905073354c0 : SUCCESS in 13m 53s\n- openstack-ansible-deploy-aio_metal-debian-bullseye https://zuul.opendev.org/t/openstack/build/ce0ec3e875f547c8bf9e0902567979cd : SUCCESS in 1h 05m 07s\n- openstack-ansible-deploy-aio_metal-centos-8-stream https://zuul.opendev.org/t/openstack/build/474e18861588476db97af04527badce2 : SUCCESS in 1h 18m 24s\n- openstack-ansible-deploy-aio_metal-ubuntu-focal https://zuul.opendev.org/t/openstack/build/273f9cd242754c10b823d72ca27e3a41 : SUCCESS in 1h 14m 26s\n- openstack-ansible-deploy-aio_metal-rockylinux-8 https://zuul.opendev.org/t/openstack/build/5f4b1921d98b47cca27cfb690032b718 : SUCCESS in 1h 21m 47s\n- openstack-ansible-upgrade-aio_metal-centos-8-stream https://zuul.opendev.org/t/openstack/build/728d1c54dda74fcbb773e34b5f718c67 : SUCCESS in 1h 52m 11s (non-voting)\n- openstack-ansible-upgrade-aio_metal-ubuntu-focal https://zuul.opendev.org/t/openstack/build/68736964899144e0ae7a492778317e37 : SUCCESS in 1h 55m 01s\n- openstack-tox-docs https://zuul.opendev.org/t/openstack/build/6bc4d70d2bc7490cba47f3fdfe478752 : SUCCESS in 4m 27s","accounts_in_message":[],"_revision_number":2},{"id":"75d46dd0f26b47f795e4a3e9df3351e7d7d56b94","author":{"_account_id":31542,"name":"Andrew Bonney","email":"andrew.bonney@bbc.co.uk","username":"andrewbonney"},"date":"2022-04-04 15:07:21.000000000","message":"Patch Set 2: Code-Review+2 Workflow+1","accounts_in_message":[],"_revision_number":2},{"id":"64653e99e8bc7cf7082e2a575cf8b07a403dadb1","tag":"autogenerated:zuul:gate","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2022-04-04 15:09:48.000000000","message":"Patch Set 2: -Verified\n\nStarting gate jobs.","accounts_in_message":[],"_revision_number":2},{"id":"ef8761f5007e880790d4c0cb0440d538656c8f76","tag":"autogenerated:zuul:gate","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2022-04-04 16:59:54.000000000","message":"Patch Set 2: Verified+2\n\nBuild succeeded (gate pipeline).\n\n- openstack-ansible-linters-ubuntu-focal https://zuul.opendev.org/t/openstack/build/fdf93817a5434493985b311c0566acde : SUCCESS in 13m 17s\n- openstack-ansible-deploy-aio_metal-debian-bullseye https://zuul.opendev.org/t/openstack/build/f1b8d32d155242a692835ff46387840f : SUCCESS in 1h 18m 59s\n- openstack-ansible-deploy-aio_metal-centos-8-stream https://zuul.opendev.org/t/openstack/build/eed1564700cd4a2aa68ab8c47f407871 : SUCCESS in 1h 45m 24s\n- openstack-ansible-deploy-aio_metal-ubuntu-focal https://zuul.opendev.org/t/openstack/build/647d3f5ae85b4dfd8cb20fa547a5e0e9 : SUCCESS in 1h 24m 55s\n- openstack-ansible-deploy-aio_metal-rockylinux-8 https://zuul.opendev.org/t/openstack/build/e7b66ddade874a9e814d919baf030b98 : SUCCESS in 1h 21m 26s\n- openstack-tox-docs https://zuul.opendev.org/t/openstack/build/8dcc4e96d95e4617b7c048d88643bdb1 : SUCCESS in 4m 52s","accounts_in_message":[],"_revision_number":2},{"id":"fd70acd6f00bfa8ff62061d039e3d08d0db1b15f","tag":"autogenerated:gerrit:merged","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2022-04-04 16:59:54.000000000","message":"Change has been successfully merged","accounts_in_message":[],"_revision_number":2},{"id":"e3408e97169d1fa640a8c170fb9bff90a4226600","tag":"autogenerated:zuul:promote","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2022-04-04 17:01:18.000000000","message":"Patch Set 2:\n\nBuild succeeded (promote pipeline).\n\n- promote-openstack-tox-docs https://zuul.opendev.org/t/openstack/build/afa8933eeceb47a3a9d9105e54dc2cb6 : SUCCESS in 58s","accounts_in_message":[],"_revision_number":2}],"current_revision_number":2,"current_revision":"621aee519c2f6a9866aca470cca0950366f8b5c8","revisions":{"e24b94a2c4f18439dd50db211ff89c6c4122d65d":{"kind":"REWORK","_number":1,"created":"2022-03-29 17:17:45.000000000","uploader":{"_account_id":11131,"name":"OpenStack Proposal Bot","username":"proposal-bot","tags":["SERVICE_USER"]},"ref":"refs/changes/97/835697/1","fetch":{"anonymous http":{"url":"https://review.opendev.org/openstack/openstack-ansible-os_adjutant","ref":"refs/changes/97/835697/1","commands":{"Checkout":"git fetch https://review.opendev.org/openstack/openstack-ansible-os_adjutant refs/changes/97/835697/1 \u0026\u0026 git checkout FETCH_HEAD","Cherry Pick":"git fetch https://review.opendev.org/openstack/openstack-ansible-os_adjutant refs/changes/97/835697/1 \u0026\u0026 git cherry-pick FETCH_HEAD","Format Patch":"git fetch https://review.opendev.org/openstack/openstack-ansible-os_adjutant refs/changes/97/835697/1 \u0026\u0026 git format-patch -1 --stdout FETCH_HEAD","Pull":"git pull https://review.opendev.org/openstack/openstack-ansible-os_adjutant refs/changes/97/835697/1"}}},"commit":{"parents":[{"commit":"64aa332ac6e33ea548cd41a2775b9bffa3168952","subject":"Remove static parameter for import","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/openstack-ansible-os_adjutant/commit/64aa332ac6e33ea548cd41a2775b9bffa3168952"}]}],"author":{"name":"OpenStack Proposal Bot","email":"openstack-infra@lists.openstack.org","date":"2022-03-29 17:17:44.000000000","tz":0},"committer":{"name":"OpenStack Proposal Bot","email":"openstack-infra@lists.openstack.org","date":"2022-03-29 17:17:44.000000000","tz":0},"subject":"Updated from OpenStack Ansible Tests","message":"Updated from OpenStack Ansible Tests\n\nChange-Id: Ifedfd0e702c91820337c0e61b7d780bab18f4892\n","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/openstack-ansible-os_adjutant/commit/e24b94a2c4f18439dd50db211ff89c6c4122d65d"}],"resolve_conflicts_web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/openstack-ansible-os_adjutant/commit/e24b94a2c4f18439dd50db211ff89c6c4122d65d"}]},"branch":"refs/heads/master"},"621aee519c2f6a9866aca470cca0950366f8b5c8":{"kind":"TRIVIAL_REBASE","_number":2,"created":"2022-04-04 09:09:52.000000000","uploader":{"_account_id":25023,"name":"Jonathan Rosser","email":"jonathan.rosser@rd.bbc.co.uk","username":"jrosser"},"ref":"refs/changes/97/835697/2","fetch":{"anonymous http":{"url":"https://review.opendev.org/openstack/openstack-ansible-os_adjutant","ref":"refs/changes/97/835697/2","commands":{"Checkout":"git fetch https://review.opendev.org/openstack/openstack-ansible-os_adjutant refs/changes/97/835697/2 \u0026\u0026 git checkout FETCH_HEAD","Cherry Pick":"git fetch https://review.opendev.org/openstack/openstack-ansible-os_adjutant refs/changes/97/835697/2 \u0026\u0026 git cherry-pick FETCH_HEAD","Format Patch":"git fetch https://review.opendev.org/openstack/openstack-ansible-os_adjutant refs/changes/97/835697/2 \u0026\u0026 git format-patch -1 --stdout FETCH_HEAD","Pull":"git pull https://review.opendev.org/openstack/openstack-ansible-os_adjutant refs/changes/97/835697/2"}}},"commit":{"parents":[{"commit":"1fd359764330c2131e4de3f5d992ac5cb8522cc7","subject":"Cleanup setup.py config","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/openstack-ansible-os_adjutant/commit/1fd359764330c2131e4de3f5d992ac5cb8522cc7"}]}],"author":{"name":"OpenStack Proposal Bot","email":"openstack-infra@lists.openstack.org","date":"2022-03-29 17:17:44.000000000","tz":0},"committer":{"name":"Jonathan Rosser","email":"jonathan.rosser@rd.bbc.co.uk","date":"2022-04-04 09:09:52.000000000","tz":0},"subject":"Updated from OpenStack Ansible Tests","message":"Updated from OpenStack Ansible Tests\n\nChange-Id: Ifedfd0e702c91820337c0e61b7d780bab18f4892\n","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/openstack-ansible-os_adjutant/commit/621aee519c2f6a9866aca470cca0950366f8b5c8"}],"resolve_conflicts_web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/openstack-ansible-os_adjutant/commit/621aee519c2f6a9866aca470cca0950366f8b5c8"}]},"branch":"refs/heads/master","description":"Rebase"}},"requirements":[],"submit_records":[{"status":"CLOSED","labels":[{"label":"Verified","status":"OK","applied_by":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]}},{"label":"Code-Review","status":"OK","applied_by":{"_account_id":31542,"name":"Andrew Bonney","email":"andrew.bonney@bbc.co.uk","username":"andrewbonney"}},{"label":"Workflow","status":"OK","applied_by":{"_account_id":31542,"name":"Andrew Bonney","email":"andrew.bonney@bbc.co.uk","username":"andrewbonney"}},{"label":"Backport-Candidate","status":"MAY"}]}],"submit_requirements":[{"name":"Verified","status":"SATISFIED","is_legacy":true,"submittability_expression_result":{"expression":"label:Verified\u003dCustom-Rule","fulfilled":true,"status":"PASS","passing_atoms":["label:Verified\u003dCustom-Rule"],"failing_atoms":[],"atom_explanations":{}}},{"name":"Workflow","status":"SATISFIED","is_legacy":true,"submittability_expression_result":{"expression":"label:Workflow\u003dCustom-Rule","fulfilled":true,"status":"PASS","passing_atoms":["label:Workflow\u003dCustom-Rule"],"failing_atoms":[],"atom_explanations":{}}},{"name":"Code-Review","status":"SATISFIED","is_legacy":true,"submittability_expression_result":{"expression":"label:Code-Review\u003dCustom-Rule","fulfilled":true,"status":"PASS","passing_atoms":["label:Code-Review\u003dCustom-Rule"],"failing_atoms":[],"atom_explanations":{}}}]}
