)]}'
{"id":"openstack%2Fopenstack-ansible-os_adjutant~879892","triplet_id":"openstack%2Fopenstack-ansible-os_adjutant~master~Ie5c4f5528a25d0a13c12749c1d7af09377792342","project":"openstack/openstack-ansible-os_adjutant","branch":"master","attention_set":{},"removed_from_attention_set":{"28619":{"account":{"_account_id":28619,"name":"Dmitriy Rabotyagov","email":"noonedeadpunk@gmail.com","username":"noonedeadpunk"},"last_update":"2023-04-22 22:32:17.000000000","reason":"Change was submitted"}},"hashtags":[],"change_id":"Ie5c4f5528a25d0a13c12749c1d7af09377792342","subject":"Switch back to use constrained Django version","status":"MERGED","created":"2023-04-07 12:22:00.000000000","updated":"2023-04-22 22:33:14.000000000","submitted":"2023-04-22 22:32:17.000000000","submitter":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"total_comment_count":3,"unresolved_comment_count":0,"has_review_started":true,"submission_id":"879892","meta_rev_id":"1ee840a23e6a9fd320df507a17b10e0f769b5673","_number":879892,"virtual_id_number":879892,"owner":{"_account_id":28619,"name":"Dmitriy Rabotyagov","email":"noonedeadpunk@gmail.com","username":"noonedeadpunk"},"actions":{},"labels":{"Verified":{"approved":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"all":[{"value":0,"_account_id":32666,"name":"Damian Dąbrowski","email":"damian@dabrowski.cloud","username":"ddabrowski"},{"tag":"autogenerated:zuul:gate","value":2,"date":"2023-04-22 22:32:16.000000000","permitted_voting_range":{"min":2,"max":2},"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},{"value":0,"_account_id":25023,"name":"Jonathan Rosser","email":"jonathan.rosser@rd.bbc.co.uk","username":"jrosser"}],"values":{"-2":"Fails","-1":"Doesn\u0027t seem to work"," 0":"No score","+1":"Works for me","+2":"Verified"},"description":"","default_value":0,"optional":true},"Code-Review":{"approved":{"_account_id":32666,"name":"Damian Dąbrowski","email":"damian@dabrowski.cloud","username":"ddabrowski"},"all":[{"value":2,"date":"2023-04-21 07:05:29.000000000","permitted_voting_range":{"min":2,"max":2},"_account_id":32666,"name":"Damian Dąbrowski","email":"damian@dabrowski.cloud","username":"ddabrowski"},{"value":0,"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},{"value":2,"date":"2023-04-22 20:36:54.000000000","permitted_voting_range":{"min":2,"max":2},"_account_id":25023,"name":"Jonathan Rosser","email":"jonathan.rosser@rd.bbc.co.uk","username":"jrosser"}],"values":{"-2":"Do not merge","-1":"This patch needs further work before it can be merged"," 0":"No score","+1":"Looks good to me, but someone else must approve","+2":"Looks good to me (core reviewer)"},"description":"","default_value":0,"optional":true},"Workflow":{"approved":{"_account_id":25023,"name":"Jonathan Rosser","email":"jonathan.rosser@rd.bbc.co.uk","username":"jrosser"},"all":[{"value":0,"_account_id":32666,"name":"Damian Dąbrowski","email":"damian@dabrowski.cloud","username":"ddabrowski"},{"value":0,"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},{"value":1,"date":"2023-04-22 20:36:54.000000000","permitted_voting_range":{"min":1,"max":1},"_account_id":25023,"name":"Jonathan Rosser","email":"jonathan.rosser@rd.bbc.co.uk","username":"jrosser"}],"values":{"-1":"Work in progress"," 0":"Ready for reviews","+1":"Approved"},"description":"","default_value":0,"optional":true},"Backport-Candidate":{"all":[{"value":0,"_account_id":32666,"name":"Damian Dąbrowski","email":"damian@dabrowski.cloud","username":"ddabrowski"},{"value":0,"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},{"value":0,"_account_id":25023,"name":"Jonathan Rosser","email":"jonathan.rosser@rd.bbc.co.uk","username":"jrosser"}],"values":{"-1":"Not A Backport Candidate"," 0":"Backport Review Needed","+1":"Proposed Backport"},"description":"","default_value":0,"optional":true}},"removable_reviewers":[],"reviewers":{"REVIEWER":[{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},{"_account_id":25023,"name":"Jonathan Rosser","email":"jonathan.rosser@rd.bbc.co.uk","username":"jrosser"},{"_account_id":32666,"name":"Damian Dąbrowski","email":"damian@dabrowski.cloud","username":"ddabrowski"}]},"pending_reviewers":{},"reviewer_updates":[{"updated":"2023-04-07 13:47:10.000000000","updated_by":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"reviewer":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"state":"REVIEWER"},{"updated":"2023-04-13 18:51:18.000000000","updated_by":{"_account_id":32666,"name":"Damian Dąbrowski","email":"damian@dabrowski.cloud","username":"ddabrowski"},"reviewer":{"_account_id":32666,"name":"Damian Dąbrowski","email":"damian@dabrowski.cloud","username":"ddabrowski"},"state":"CC"},{"updated":"2023-04-21 07:05:29.000000000","updated_by":{"_account_id":32666,"name":"Damian Dąbrowski","email":"damian@dabrowski.cloud","username":"ddabrowski"},"reviewer":{"_account_id":32666,"name":"Damian Dąbrowski","email":"damian@dabrowski.cloud","username":"ddabrowski"},"state":"REVIEWER"},{"updated":"2023-04-22 20:36:54.000000000","updated_by":{"_account_id":25023,"name":"Jonathan Rosser","email":"jonathan.rosser@rd.bbc.co.uk","username":"jrosser"},"reviewer":{"_account_id":25023,"name":"Jonathan Rosser","email":"jonathan.rosser@rd.bbc.co.uk","username":"jrosser"},"state":"REVIEWER"}],"messages":[{"id":"f7bd16832b3e9ad817e14e8df9e9c1cd08c527ef","tag":"autogenerated:gerrit:newPatchSet","author":{"_account_id":28619,"name":"Dmitriy Rabotyagov","email":"noonedeadpunk@gmail.com","username":"noonedeadpunk"},"date":"2023-04-07 12:22:00.000000000","message":"Uploaded patch set 1.","accounts_in_message":[],"_revision_number":1},{"id":"2bbc71ddf197ec9db1881a91ca9427a87021cf08","tag":"autogenerated:zuul:check","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2023-04-07 13:47:10.000000000","message":"Patch Set 1: Verified-1\n\nBuild failed (check pipeline).  For information on how to proceed, see\nhttps://docs.opendev.org/opendev/infra-manual/latest/developers.html#automated-testing\n\nhttps://zuul.opendev.org/t/openstack/buildset/47ab1f5a72e34eab80f863b63925f868\n\n- openstack-ansible-linters-ubuntu-focal https://zuul.opendev.org/t/openstack/build/d651ffdbfa1c4cb0b7e11823ce9cf5d3 : SUCCESS in 16m 39s\n- openstack-ansible-deploy-aio_metal-debian-bullseye https://zuul.opendev.org/t/openstack/build/af100a57cdeb46bc9d4269a6acb1db05 : FAILURE in 1h 13m 35s\n- openstack-ansible-deploy-aio_metal-centos-9-stream https://zuul.opendev.org/t/openstack/build/ec9ef192cc1943a083b353f209fdc522 : FAILURE in 26m 44s\n- openstack-ansible-deploy-aio_metal-rockylinux-9 https://zuul.opendev.org/t/openstack/build/210acad375984f68b573c3f3a6cd77d9 : FAILURE in 1h 17m 51s\n- openstack-ansible-deploy-aio_metal-ubuntu-focal https://zuul.opendev.org/t/openstack/build/03ee652b4aca4385acc263bc54742b47 : FAILURE in 1h 20m 24s\n- openstack-ansible-deploy-aio_metal-ubuntu-jammy https://zuul.opendev.org/t/openstack/build/5800214547a2457a88847d411ff021b3 : FAILURE in 1h 10m 40s\n- openstack-ansible-upgrade-aio_metal-centos-9-stream https://zuul.opendev.org/t/openstack/build/c2d662fc51404180b20587028e343334 : FAILURE in 21m 03s (non-voting)\n- openstack-ansible-upgrade-aio_metal-ubuntu-focal https://zuul.opendev.org/t/openstack/build/dec69303fd7e4fc28ee580a8ef5c1a3a : FAILURE in 58m 17s\n- openstack-tox-docs https://zuul.opendev.org/t/openstack/build/d293d6a8fe244dd08a56a55aa05c0fa2 : SUCCESS in 5m 48s","accounts_in_message":[],"_revision_number":1},{"id":"2b9dd6096c24a509ab3f6b62e2baec67c45ff4cf","tag":"autogenerated:gerrit:newPatchSet","author":{"_account_id":28619,"name":"Dmitriy Rabotyagov","email":"noonedeadpunk@gmail.com","username":"noonedeadpunk"},"date":"2023-04-10 08:42:41.000000000","message":"Patch Set 2: Published edit on patch set 1.\n\nOutdated Votes:\n* Verified-1\n","accounts_in_message":[],"_revision_number":2},{"id":"7167bea07b1efe64a12f9b06418fac9a2d296165","tag":"autogenerated:zuul:check","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2023-04-10 10:23:37.000000000","message":"Patch Set 2: Verified-1\n\nBuild failed (check pipeline).  For information on how to proceed, see\nhttps://docs.opendev.org/opendev/infra-manual/latest/developers.html#automated-testing\n\nhttps://zuul.opendev.org/t/openstack/buildset/f2f8bc5424014beb9108f913c2762947\n\n- openstack-ansible-linters-ubuntu-focal https://zuul.opendev.org/t/openstack/build/dda8e7ba00ff484180b0544b404ad755 : SUCCESS in 11m 58s\n- openstack-ansible-deploy-aio_metal-debian-bullseye https://zuul.opendev.org/t/openstack/build/70d9b23b3b7d46f58ef6b083548935e6 : FAILURE in 1h 21m 02s\n- openstack-ansible-deploy-aio_metal-centos-9-stream https://zuul.opendev.org/t/openstack/build/761547e68c484ae5a33f214380c6309a : FAILURE in 1h 11m 22s (non-voting)\n- openstack-ansible-deploy-aio_metal-rockylinux-9 https://zuul.opendev.org/t/openstack/build/fd233c6f9ae24bc182672a437d8e313b : FAILURE in 1h 33m 42s\n- openstack-ansible-deploy-aio_metal-ubuntu-focal https://zuul.opendev.org/t/openstack/build/20df72dcca704a7fbcec27128e898901 : FAILURE in 1h 25m 11s\n- openstack-ansible-deploy-aio_metal-ubuntu-jammy https://zuul.opendev.org/t/openstack/build/8b44665d955e4d3e9c55bf0772f68052 : FAILURE in 1h 09m 19s\n- openstack-ansible-upgrade-aio_metal-rockylinux-9 https://zuul.opendev.org/t/openstack/build/99fcca1dbe554c92814faa6d41a3348d : FAILURE in 1h 05m 48s\n- openstack-ansible-upgrade-aio_metal-ubuntu-focal https://zuul.opendev.org/t/openstack/build/757b15e32ec44ecbbc4dd9daf0f88d82 : FAILURE in 1h 08m 24s\n- openstack-tox-docs https://zuul.opendev.org/t/openstack/build/e83604b517664fb481c224bce87d3c90 : SUCCESS in 5m 22s","accounts_in_message":[],"_revision_number":2},{"id":"1d19f0c204414ff7beca910a66f3f7f1ab9a31fe","tag":"autogenerated:gerrit:newPatchSet","author":{"_account_id":28619,"name":"Dmitriy Rabotyagov","email":"noonedeadpunk@gmail.com","username":"noonedeadpunk"},"date":"2023-04-10 13:21:26.000000000","message":"Patch Set 3: Published edit on patch set 2.\n\nOutdated Votes:\n* Verified-1\n","accounts_in_message":[],"_revision_number":3},{"id":"aa46cd592089c6c4d2841608fa77bf9519afad6a","tag":"autogenerated:gerrit:newPatchSet","author":{"_account_id":28619,"name":"Dmitriy Rabotyagov","email":"noonedeadpunk@gmail.com","username":"noonedeadpunk"},"date":"2023-04-10 13:22:31.000000000","message":"Patch Set 4: Commit message was updated.","accounts_in_message":[],"_revision_number":4},{"id":"d449f6a3f1ef40640dd6ae670a948eb3e5b79935","tag":"autogenerated:zuul:check","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2023-04-10 15:05:29.000000000","message":"Patch Set 4: Verified-1\n\nBuild failed (check pipeline).  For information on how to proceed, see\nhttps://docs.opendev.org/opendev/infra-manual/latest/developers.html#automated-testing\n\nhttps://zuul.opendev.org/t/openstack/buildset/66f33afc84354aada29ab3ff7d8292ca\n\n- openstack-ansible-linters-ubuntu-focal https://zuul.opendev.org/t/openstack/build/9bf7d433b64c4c0ca5ffb4c51fc83fbb : SUCCESS in 10m 02s\n- openstack-ansible-deploy-aio_metal-debian-bullseye https://zuul.opendev.org/t/openstack/build/4f52d0eb5a2d4ccfafbe2dc1b3fac066 : SUCCESS in 1h 33m 55s\n- openstack-ansible-deploy-aio_metal-centos-9-stream https://zuul.opendev.org/t/openstack/build/55a20c75f92747cea1c6db3f3e7c834e : SUCCESS in 1h 30m 45s (non-voting)\n- openstack-ansible-deploy-aio_metal-rockylinux-9 https://zuul.opendev.org/t/openstack/build/26d3d4a87b5f4af7911259330007669a : SUCCESS in 1h 35m 52s\n- openstack-ansible-deploy-aio_metal-ubuntu-focal https://zuul.opendev.org/t/openstack/build/dba398ae4e3644a99834749afc9d0dad : SUCCESS in 1h 33m 46s\n- openstack-ansible-deploy-aio_metal-ubuntu-jammy https://zuul.opendev.org/t/openstack/build/1900418d6d734b1d92fbd3df15b2f899 : SUCCESS in 1h 26m 51s\n- openstack-ansible-upgrade-aio_metal-rockylinux-9 https://zuul.opendev.org/t/openstack/build/fda8dd8f7f0c4e32905685e33f7222e0 : FAILURE in 1h 08m 36s\n- openstack-ansible-upgrade-aio_metal-ubuntu-focal https://zuul.opendev.org/t/openstack/build/f28453cc7ffc4c0b82e41adc61858b02 : FAILURE in 57m 28s\n- openstack-tox-docs https://zuul.opendev.org/t/openstack/build/a644f40845fa4eccbfe83b6f2a8f4582 : SUCCESS in 4m 28s","accounts_in_message":[],"_revision_number":4},{"id":"713057700010968099cfaf3ed508027ff0aa2e98","tag":"autogenerated:gerrit:newPatchSet","author":{"_account_id":28619,"name":"Dmitriy Rabotyagov","email":"noonedeadpunk@gmail.com","username":"noonedeadpunk"},"date":"2023-04-10 16:52:59.000000000","message":"Patch Set 5: Published edit on patch set 4.\n\nOutdated Votes:\n* Verified-1\n","accounts_in_message":[],"_revision_number":5},{"id":"8c1b1684b10ef281bb96d5e2edf282d56af91005","tag":"autogenerated:gerrit:newPatchSet","author":{"_account_id":28619,"name":"Dmitriy Rabotyagov","email":"noonedeadpunk@gmail.com","username":"noonedeadpunk"},"date":"2023-04-10 16:53:07.000000000","message":"Patch Set 6: Commit message was updated.","accounts_in_message":[],"_revision_number":6},{"id":"f6030f624a15a51c4264462e6a3951b15c553ff9","tag":"autogenerated:gerrit:newPatchSet","author":{"_account_id":28619,"name":"Dmitriy Rabotyagov","email":"noonedeadpunk@gmail.com","username":"noonedeadpunk"},"date":"2023-04-10 16:54:32.000000000","message":"Patch Set 7: Published edit on patch set 6.","accounts_in_message":[],"_revision_number":7},{"id":"e6376f6e63fca1fb3d26a14a34e88db1f085f5d1","tag":"autogenerated:zuul:check","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2023-04-10 18:44:39.000000000","message":"Patch Set 7: Verified-1\n\nBuild failed (check pipeline).  For information on how to proceed, see\nhttps://docs.opendev.org/opendev/infra-manual/latest/developers.html#automated-testing\n\nhttps://zuul.opendev.org/t/openstack/buildset/b37d5e9ed8d0490bb3ef21ce71d94838\n\n- openstack-ansible-linters-ubuntu-focal https://zuul.opendev.org/t/openstack/build/cee10018cdbe4ee28cdb3a728397605b : SUCCESS in 12m 56s\n- openstack-ansible-deploy-aio_metal-debian-bullseye https://zuul.opendev.org/t/openstack/build/3371b728f02b45e3a27b28299b9a4def : SUCCESS in 1h 27m 27s\n- openstack-ansible-deploy-aio_metal-centos-9-stream https://zuul.opendev.org/t/openstack/build/cbb994bd222b4387a8493d63666cba83 : SUCCESS in 1h 24m 58s (non-voting)\n- openstack-ansible-deploy-aio_metal-rockylinux-9 https://zuul.opendev.org/t/openstack/build/df383a0eb0c041d7888f843f4bcc4fd1 : SUCCESS in 1h 36m 03s\n- openstack-ansible-deploy-aio_metal-ubuntu-focal https://zuul.opendev.org/t/openstack/build/1da9f13056534b5bbf5c00fd22314c6a : SUCCESS in 1h 27m 37s\n- openstack-ansible-deploy-aio_metal-ubuntu-jammy https://zuul.opendev.org/t/openstack/build/2a7f69dbd3db4d898e2e1da696942ef7 : SUCCESS in 1h 36m 17s\n- openstack-ansible-upgrade-aio_metal-rockylinux-9 https://zuul.opendev.org/t/openstack/build/6649ed4a6f714f8db730fe4182cbabb3 : FAILURE in 53m 58s\n- openstack-ansible-upgrade-aio_metal-ubuntu-focal https://zuul.opendev.org/t/openstack/build/2d38605568d746d2bbfa995b50d9baf7 : FAILURE in 57m 51s\n- openstack-tox-docs https://zuul.opendev.org/t/openstack/build/b76083f1f1ae4836b0c684f25525cfbe : SUCCESS in 4m 41s","accounts_in_message":[],"_revision_number":7},{"id":"009cc9f2e2564ae891c036e378496d8940342d68","author":{"_account_id":32666,"name":"Damian Dąbrowski","email":"damian@dabrowski.cloud","username":"ddabrowski"},"date":"2023-04-13 18:51:18.000000000","message":"Patch Set 7:\n\n(1 comment)","accounts_in_message":[],"_revision_number":7},{"id":"8a32bf31da2b0e01cc96cb8898ecedd4425cddb6","author":{"_account_id":32666,"name":"Damian Dąbrowski","email":"damian@dabrowski.cloud","username":"ddabrowski"},"date":"2023-04-13 18:52:14.000000000","message":"Patch Set 7:\n\n(1 comment)","accounts_in_message":[],"_revision_number":7},{"id":"1ed0ef38712d4dd898ca64b8ea90769a8fe0d1bd","tag":"autogenerated:zuul:check","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2023-04-13 20:56:25.000000000","message":"Patch Set 7:\n\nBuild failed (check pipeline).  For information on how to proceed, see\nhttps://docs.opendev.org/opendev/infra-manual/latest/developers.html#automated-testing\n\nhttps://zuul.opendev.org/t/openstack/buildset/9814b590ada9464b97c57fa80f694e23\n\n- openstack-ansible-linters-ubuntu-focal https://zuul.opendev.org/t/openstack/build/b1f01e8a268646f19df9511dd32d7381 : SUCCESS in 13m 10s\n- openstack-ansible-deploy-aio_metal-debian-bullseye https://zuul.opendev.org/t/openstack/build/78b5e909f117453fba0264cc10385bdf : SUCCESS in 1h 32m 08s\n- openstack-ansible-deploy-aio_metal-centos-9-stream https://zuul.opendev.org/t/openstack/build/f09be1fc0795440f848268ec2d174a85 : SUCCESS in 1h 39m 39s (non-voting)\n- openstack-ansible-deploy-aio_metal-rockylinux-9 https://zuul.opendev.org/t/openstack/build/e73c0a7f875b443a86d5cf54a54a9690 : SUCCESS in 1h 44m 14s\n- openstack-ansible-deploy-aio_metal-ubuntu-focal https://zuul.opendev.org/t/openstack/build/842a161f0f6f4ffe937d201cdc510d2d : SUCCESS in 1h 31m 51s\n- openstack-ansible-deploy-aio_metal-ubuntu-jammy https://zuul.opendev.org/t/openstack/build/6b1519813478465ea98cbf2fdb9921b2 : SUCCESS in 1h 34m 26s\n- openstack-ansible-upgrade-aio_metal-rockylinux-9 https://zuul.opendev.org/t/openstack/build/e343bbda44a94f77ae878047cd875f29 : FAILURE in 1h 04m 30s\n- openstack-ansible-upgrade-aio_metal-ubuntu-focal https://zuul.opendev.org/t/openstack/build/b2ffb5c7c94d45959dab0ffe55ab4dbe : FAILURE in 1h 01m 15s\n- openstack-tox-docs https://zuul.opendev.org/t/openstack/build/ab2a26f0e1524aacb09cd2709358f865 : SUCCESS in 5m 13s","accounts_in_message":[],"_revision_number":7},{"id":"10042661dd2f58c71cda918485ed65a568aac236","author":{"_account_id":32666,"name":"Damian Dąbrowski","email":"damian@dabrowski.cloud","username":"ddabrowski"},"date":"2023-04-20 20:45:29.000000000","message":"Patch Set 7:\n\n(1 comment)","accounts_in_message":[],"_revision_number":7},{"id":"46a904263f9d3b38c47af73eea4ee47ae85fb095","tag":"autogenerated:zuul:check","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2023-04-20 23:39:03.000000000","message":"Patch Set 7: Verified+1\n\nBuild succeeded (check pipeline).\nhttps://zuul.opendev.org/t/openstack/buildset/d83833e54caa4c7885edc6faa51a4e7d\n\n- openstack-ansible-linters-ubuntu-focal https://zuul.opendev.org/t/openstack/build/84e5b57284a7453abb56b77bd8b5c32f : SUCCESS in 11m 06s\n- openstack-ansible-deploy-aio_metal-debian-bullseye https://zuul.opendev.org/t/openstack/build/83eb18c7e8c84ecf857a874b137b56c2 : SUCCESS in 1h 27m 46s\n- openstack-ansible-deploy-aio_metal-centos-9-stream https://zuul.opendev.org/t/openstack/build/770f5b89782e4282b690311a554535f8 : SUCCESS in 1h 30m 16s (non-voting)\n- openstack-ansible-deploy-aio_metal-rockylinux-9 https://zuul.opendev.org/t/openstack/build/59f0df91718c48c882b06bbdf8c977bf : SUCCESS in 1h 37m 54s\n- openstack-ansible-deploy-aio_metal-ubuntu-focal https://zuul.opendev.org/t/openstack/build/d0b9a82db6524409a78d08031164200c : SUCCESS in 1h 40m 39s\n- openstack-ansible-deploy-aio_metal-ubuntu-jammy https://zuul.opendev.org/t/openstack/build/f0dc541bd36647be820b48e5387dac9a : SUCCESS in 1h 34m 50s\n- openstack-ansible-upgrade-aio_metal-rockylinux-9 https://zuul.opendev.org/t/openstack/build/57905de08cf746cd827cdc537d81babf : SUCCESS in 2h 46m 35s\n- openstack-ansible-upgrade-aio_metal-ubuntu-focal https://zuul.opendev.org/t/openstack/build/81365aeaa53d422c9172096450cb8a86 : SUCCESS in 2h 13m 34s\n- openstack-tox-docs https://zuul.opendev.org/t/openstack/build/5718a75510ee4636ac3bf5b40c009513 : SUCCESS in 4m 52s","accounts_in_message":[],"_revision_number":7},{"id":"c0c6f26257eb433bb7e705ade8f926bd8ec81b6f","author":{"_account_id":32666,"name":"Damian Dąbrowski","email":"damian@dabrowski.cloud","username":"ddabrowski"},"date":"2023-04-21 07:05:29.000000000","message":"Patch Set 7: Code-Review+2","accounts_in_message":[],"_revision_number":7},{"id":"f9bd8c980f12b0d62fab1fe7a1d0438ad44fca55","author":{"_account_id":25023,"name":"Jonathan Rosser","email":"jonathan.rosser@rd.bbc.co.uk","username":"jrosser"},"date":"2023-04-22 20:36:54.000000000","message":"Patch Set 7: Code-Review+2 Workflow+1","accounts_in_message":[],"_revision_number":7},{"id":"4d17aecaa009ac3edb641b7032c2831592b5dce4","tag":"autogenerated:zuul:gate","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2023-04-22 20:40:01.000000000","message":"Patch Set 7: -Verified\n\nStarting gate jobs.","accounts_in_message":[],"_revision_number":7},{"id":"9fd3782e987b7f5fab097ebed0a5c7626183bc91","tag":"autogenerated:zuul:gate","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2023-04-22 22:32:16.000000000","message":"Patch Set 7: Verified+2\n\nBuild succeeded (gate pipeline).\nhttps://zuul.opendev.org/t/openstack/buildset/7ade5c7c1b3e431baeb665713e5fc348\n\n- openstack-ansible-linters-ubuntu-focal https://zuul.opendev.org/t/openstack/build/a09ff8f130ff4736a698279eec5de182 : SUCCESS in 10m 59s\n- openstack-ansible-deploy-aio_metal-debian-bullseye https://zuul.opendev.org/t/openstack/build/51573a2027fd4f6bb7fc1edf8ed566a2 : SUCCESS in 1h 09m 33s\n- openstack-ansible-deploy-aio_metal-rockylinux-9 https://zuul.opendev.org/t/openstack/build/4ea942bffe3d426982e8177e9c723a67 : SUCCESS in 1h 45m 28s\n- openstack-ansible-deploy-aio_metal-ubuntu-jammy https://zuul.opendev.org/t/openstack/build/9b9b018fc068433f8fa722525c25de66 : SUCCESS in 1h 15m 02s\n- openstack-tox-docs https://zuul.opendev.org/t/openstack/build/8e73636563ba46da8b76238fd2023166 : SUCCESS in 3m 59s","accounts_in_message":[],"_revision_number":7},{"id":"ac0b304ef60389c68354b2543a56e1ca928fb08e","tag":"autogenerated:gerrit:merged","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2023-04-22 22:32:17.000000000","message":"Change has been successfully merged","accounts_in_message":[],"_revision_number":7},{"id":"1ee840a23e6a9fd320df507a17b10e0f769b5673","tag":"autogenerated:zuul:promote","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2023-04-22 22:33:14.000000000","message":"Patch Set 7:\n\nBuild succeeded (promote pipeline).\nhttps://zuul.opendev.org/t/openstack/buildset/c011e8fa56bb429fa6a3772cc8f92bde\n\n- promote-openstack-tox-docs https://zuul.opendev.org/t/openstack/build/9f53f14201ca4726ae6b22bc2f8b6885 : SUCCESS in 41s","accounts_in_message":[],"_revision_number":7}],"current_revision_number":7,"current_revision":"7444c804a9de8b71d568f63c35f47a5dda697267","revisions":{"d6c4d7ef412e99315cbe69ab0c92589cc10d187f":{"kind":"REWORK","_number":1,"created":"2023-04-07 12:22:00.000000000","uploader":{"_account_id":28619,"name":"Dmitriy Rabotyagov","email":"noonedeadpunk@gmail.com","username":"noonedeadpunk"},"ref":"refs/changes/92/879892/1","fetch":{"anonymous http":{"url":"https://review.opendev.org/openstack/openstack-ansible-os_adjutant","ref":"refs/changes/92/879892/1","commands":{"Checkout":"git fetch https://review.opendev.org/openstack/openstack-ansible-os_adjutant refs/changes/92/879892/1 \u0026\u0026 git checkout FETCH_HEAD","Cherry Pick":"git fetch https://review.opendev.org/openstack/openstack-ansible-os_adjutant refs/changes/92/879892/1 \u0026\u0026 git cherry-pick FETCH_HEAD","Format Patch":"git fetch https://review.opendev.org/openstack/openstack-ansible-os_adjutant refs/changes/92/879892/1 \u0026\u0026 git format-patch -1 --stdout FETCH_HEAD","Pull":"git pull https://review.opendev.org/openstack/openstack-ansible-os_adjutant refs/changes/92/879892/1"}}},"commit":{"parents":[{"commit":"7f2d4a20736b63116c8ee03e10f0e2370544195c","subject":"Update master for stable/zed","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/openstack-ansible-os_adjutant/commit/7f2d4a20736b63116c8ee03e10f0e2370544195c"}]}],"author":{"name":"Dmitriy Rabotyagov","email":"dmitriy.rabotyagov@cleura.com","date":"2023-04-07 12:21:53.000000000","tz":120},"committer":{"name":"Dmitriy Rabotyagov","email":"dmitriy.rabotyagov@cleura.com","date":"2023-04-07 12:21:53.000000000","tz":120},"subject":"Switch back to use constrained Django version","message":"Switch back to use constrained Django version\n\nSupposedly, Adjutant has upgraded used Django version and constrained\nversion can be used now, since upper limit for Django is not defined\nanymore in project requirements [1]\n\n[1] https://opendev.org/openstack/adjutant/src/commit/df8c3e4a8b70be8697ac46f0acec9169752b4698/requirements.txt#L3\n\nChange-Id: Ie5c4f5528a25d0a13c12749c1d7af09377792342\n","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/openstack-ansible-os_adjutant/commit/d6c4d7ef412e99315cbe69ab0c92589cc10d187f"}],"resolve_conflicts_web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/openstack-ansible-os_adjutant/commit/d6c4d7ef412e99315cbe69ab0c92589cc10d187f"}]},"branch":"refs/heads/master"},"afb6689afb62b04e6d983b0f67f1ecd435cb8f00":{"kind":"REWORK","_number":2,"created":"2023-04-10 08:42:41.000000000","uploader":{"_account_id":28619,"name":"Dmitriy Rabotyagov","email":"noonedeadpunk@gmail.com","username":"noonedeadpunk"},"ref":"refs/changes/92/879892/2","fetch":{"anonymous http":{"url":"https://review.opendev.org/openstack/openstack-ansible-os_adjutant","ref":"refs/changes/92/879892/2","commands":{"Checkout":"git fetch https://review.opendev.org/openstack/openstack-ansible-os_adjutant refs/changes/92/879892/2 \u0026\u0026 git checkout FETCH_HEAD","Cherry Pick":"git fetch https://review.opendev.org/openstack/openstack-ansible-os_adjutant refs/changes/92/879892/2 \u0026\u0026 git cherry-pick FETCH_HEAD","Format Patch":"git fetch https://review.opendev.org/openstack/openstack-ansible-os_adjutant refs/changes/92/879892/2 \u0026\u0026 git format-patch -1 --stdout FETCH_HEAD","Pull":"git pull https://review.opendev.org/openstack/openstack-ansible-os_adjutant refs/changes/92/879892/2"}}},"commit":{"parents":[{"commit":"7f2d4a20736b63116c8ee03e10f0e2370544195c","subject":"Update master for stable/zed","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/openstack-ansible-os_adjutant/commit/7f2d4a20736b63116c8ee03e10f0e2370544195c"}]}],"author":{"name":"Dmitriy Rabotyagov","email":"dmitriy.rabotyagov@cleura.com","date":"2023-04-07 12:21:53.000000000","tz":120},"committer":{"name":"Dmitriy Rabotyagov","email":"noonedeadpunk@gmail.com","date":"2023-04-10 08:42:41.000000000","tz":0},"subject":"Switch back to use constrained Django version","message":"Switch back to use constrained Django version\n\nSupposedly, Adjutant has upgraded used Django version and constrained\nversion can be used now, since upper limit for Django is not defined\nanymore in project requirements [1]\n\n[1] https://opendev.org/openstack/adjutant/src/commit/df8c3e4a8b70be8697ac46f0acec9169752b4698/requirements.txt#L3\n\nChange-Id: Ie5c4f5528a25d0a13c12749c1d7af09377792342\n","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/openstack-ansible-os_adjutant/commit/afb6689afb62b04e6d983b0f67f1ecd435cb8f00"}],"resolve_conflicts_web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/openstack-ansible-os_adjutant/commit/afb6689afb62b04e6d983b0f67f1ecd435cb8f00"}]},"branch":"refs/heads/master"},"22d007fab6f00a64674b305fa30da9a4be618455":{"kind":"REWORK","_number":3,"created":"2023-04-10 13:21:26.000000000","uploader":{"_account_id":28619,"name":"Dmitriy Rabotyagov","email":"noonedeadpunk@gmail.com","username":"noonedeadpunk"},"ref":"refs/changes/92/879892/3","fetch":{"anonymous http":{"url":"https://review.opendev.org/openstack/openstack-ansible-os_adjutant","ref":"refs/changes/92/879892/3","commands":{"Checkout":"git fetch https://review.opendev.org/openstack/openstack-ansible-os_adjutant refs/changes/92/879892/3 \u0026\u0026 git checkout FETCH_HEAD","Cherry Pick":"git fetch https://review.opendev.org/openstack/openstack-ansible-os_adjutant refs/changes/92/879892/3 \u0026\u0026 git cherry-pick FETCH_HEAD","Format Patch":"git fetch https://review.opendev.org/openstack/openstack-ansible-os_adjutant refs/changes/92/879892/3 \u0026\u0026 git format-patch -1 --stdout FETCH_HEAD","Pull":"git pull https://review.opendev.org/openstack/openstack-ansible-os_adjutant refs/changes/92/879892/3"}}},"commit":{"parents":[{"commit":"7f2d4a20736b63116c8ee03e10f0e2370544195c","subject":"Update master for stable/zed","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/openstack-ansible-os_adjutant/commit/7f2d4a20736b63116c8ee03e10f0e2370544195c"}]}],"author":{"name":"Dmitriy Rabotyagov","email":"dmitriy.rabotyagov@cleura.com","date":"2023-04-07 12:21:53.000000000","tz":120},"committer":{"name":"Dmitriy Rabotyagov","email":"noonedeadpunk@gmail.com","date":"2023-04-10 13:21:26.000000000","tz":0},"subject":"Switch back to use constrained Django version","message":"Switch back to use constrained Django version\n\nSupposedly, Adjutant has upgraded used Django version and constrained\nversion can be used now, since upper limit for Django is not defined\nanymore in project requirements [1]\n\n[1] https://opendev.org/openstack/adjutant/src/commit/df8c3e4a8b70be8697ac46f0acec9169752b4698/requirements.txt#L3\n\nChange-Id: Ie5c4f5528a25d0a13c12749c1d7af09377792342\n","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/openstack-ansible-os_adjutant/commit/22d007fab6f00a64674b305fa30da9a4be618455"}],"resolve_conflicts_web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/openstack-ansible-os_adjutant/commit/22d007fab6f00a64674b305fa30da9a4be618455"}]},"branch":"refs/heads/master"},"5e712d58e3c8d2b566f0fc41f2e99cc0a9a6b798":{"kind":"NO_CODE_CHANGE","_number":4,"created":"2023-04-10 13:22:31.000000000","uploader":{"_account_id":28619,"name":"Dmitriy Rabotyagov","email":"noonedeadpunk@gmail.com","username":"noonedeadpunk"},"ref":"refs/changes/92/879892/4","fetch":{"anonymous http":{"url":"https://review.opendev.org/openstack/openstack-ansible-os_adjutant","ref":"refs/changes/92/879892/4","commands":{"Checkout":"git fetch https://review.opendev.org/openstack/openstack-ansible-os_adjutant refs/changes/92/879892/4 \u0026\u0026 git checkout FETCH_HEAD","Cherry Pick":"git fetch https://review.opendev.org/openstack/openstack-ansible-os_adjutant refs/changes/92/879892/4 \u0026\u0026 git cherry-pick FETCH_HEAD","Format Patch":"git fetch https://review.opendev.org/openstack/openstack-ansible-os_adjutant refs/changes/92/879892/4 \u0026\u0026 git format-patch -1 --stdout FETCH_HEAD","Pull":"git pull https://review.opendev.org/openstack/openstack-ansible-os_adjutant refs/changes/92/879892/4"}}},"commit":{"parents":[{"commit":"7f2d4a20736b63116c8ee03e10f0e2370544195c","subject":"Update master for stable/zed","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/openstack-ansible-os_adjutant/commit/7f2d4a20736b63116c8ee03e10f0e2370544195c"}]}],"author":{"name":"Dmitriy Rabotyagov","email":"dmitriy.rabotyagov@cleura.com","date":"2023-04-07 12:21:53.000000000","tz":120},"committer":{"name":"Dmitriy Rabotyagov","email":"noonedeadpunk@gmail.com","date":"2023-04-10 13:22:31.000000000","tz":0},"subject":"Switch back to use constrained Django version","message":"Switch back to use constrained Django version\n\nSupposedly, Adjutant has upgraded used Django version and constrained\nversion can be used now, since upper limit for Django is not defined\nanymore in project requirements [1]\n\nWe also fix ansible_facts usage with same commit to resolve\ncircular dependency.\n\n[1] https://opendev.org/openstack/adjutant/src/commit/df8c3e4a8b70be8697ac46f0acec9169752b4698/requirements.txt#L3\n\nChange-Id: Ie5c4f5528a25d0a13c12749c1d7af09377792342\n","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/openstack-ansible-os_adjutant/commit/5e712d58e3c8d2b566f0fc41f2e99cc0a9a6b798"}],"resolve_conflicts_web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/openstack-ansible-os_adjutant/commit/5e712d58e3c8d2b566f0fc41f2e99cc0a9a6b798"}]},"branch":"refs/heads/master","description":"Edit commit message"},"b0d628717ccfac04919d2bbcfb13588e4d3b1c24":{"kind":"REWORK","_number":5,"created":"2023-04-10 16:52:59.000000000","uploader":{"_account_id":28619,"name":"Dmitriy Rabotyagov","email":"noonedeadpunk@gmail.com","username":"noonedeadpunk"},"ref":"refs/changes/92/879892/5","fetch":{"anonymous http":{"url":"https://review.opendev.org/openstack/openstack-ansible-os_adjutant","ref":"refs/changes/92/879892/5","commands":{"Checkout":"git fetch https://review.opendev.org/openstack/openstack-ansible-os_adjutant refs/changes/92/879892/5 \u0026\u0026 git checkout FETCH_HEAD","Cherry Pick":"git fetch https://review.opendev.org/openstack/openstack-ansible-os_adjutant refs/changes/92/879892/5 \u0026\u0026 git cherry-pick FETCH_HEAD","Format Patch":"git fetch https://review.opendev.org/openstack/openstack-ansible-os_adjutant refs/changes/92/879892/5 \u0026\u0026 git format-patch -1 --stdout FETCH_HEAD","Pull":"git pull https://review.opendev.org/openstack/openstack-ansible-os_adjutant refs/changes/92/879892/5"}}},"commit":{"parents":[{"commit":"7f2d4a20736b63116c8ee03e10f0e2370544195c","subject":"Update master for stable/zed","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/openstack-ansible-os_adjutant/commit/7f2d4a20736b63116c8ee03e10f0e2370544195c"}]}],"author":{"name":"Dmitriy Rabotyagov","email":"dmitriy.rabotyagov@cleura.com","date":"2023-04-07 12:21:53.000000000","tz":120},"committer":{"name":"Dmitriy Rabotyagov","email":"noonedeadpunk@gmail.com","date":"2023-04-10 16:52:59.000000000","tz":0},"subject":"Switch back to use constrained Django version","message":"Switch back to use constrained Django version\n\nSupposedly, Adjutant has upgraded used Django version and constrained\nversion can be used now, since upper limit for Django is not defined\nanymore in project requirements [1]\n\nWe also fix ansible_facts usage with same commit to resolve\ncircular dependency.\n\n[1] https://opendev.org/openstack/adjutant/src/commit/df8c3e4a8b70be8697ac46f0acec9169752b4698/requirements.txt#L3\n\nChange-Id: Ie5c4f5528a25d0a13c12749c1d7af09377792342\n","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/openstack-ansible-os_adjutant/commit/b0d628717ccfac04919d2bbcfb13588e4d3b1c24"}],"resolve_conflicts_web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/openstack-ansible-os_adjutant/commit/b0d628717ccfac04919d2bbcfb13588e4d3b1c24"}]},"branch":"refs/heads/master"},"ebca0060c12c6ecdb0f4052ff3f6022d01663864":{"kind":"NO_CODE_CHANGE","_number":6,"created":"2023-04-10 16:53:07.000000000","uploader":{"_account_id":28619,"name":"Dmitriy Rabotyagov","email":"noonedeadpunk@gmail.com","username":"noonedeadpunk"},"ref":"refs/changes/92/879892/6","fetch":{"anonymous http":{"url":"https://review.opendev.org/openstack/openstack-ansible-os_adjutant","ref":"refs/changes/92/879892/6","commands":{"Checkout":"git fetch https://review.opendev.org/openstack/openstack-ansible-os_adjutant refs/changes/92/879892/6 \u0026\u0026 git checkout FETCH_HEAD","Cherry Pick":"git fetch https://review.opendev.org/openstack/openstack-ansible-os_adjutant refs/changes/92/879892/6 \u0026\u0026 git cherry-pick FETCH_HEAD","Format Patch":"git fetch https://review.opendev.org/openstack/openstack-ansible-os_adjutant refs/changes/92/879892/6 \u0026\u0026 git format-patch -1 --stdout FETCH_HEAD","Pull":"git pull https://review.opendev.org/openstack/openstack-ansible-os_adjutant refs/changes/92/879892/6"}}},"commit":{"parents":[{"commit":"7f2d4a20736b63116c8ee03e10f0e2370544195c","subject":"Update master for stable/zed","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/openstack-ansible-os_adjutant/commit/7f2d4a20736b63116c8ee03e10f0e2370544195c"}]}],"author":{"name":"Dmitriy Rabotyagov","email":"dmitriy.rabotyagov@cleura.com","date":"2023-04-07 12:21:53.000000000","tz":120},"committer":{"name":"Dmitriy Rabotyagov","email":"noonedeadpunk@gmail.com","date":"2023-04-10 16:53:07.000000000","tz":0},"subject":"Switch back to use constrained Django version","message":"Switch back to use constrained Django version\n\nSupposedly, Adjutant has upgraded used Django version and constrained\nversion can be used now, since upper limit for Django is not defined\nanymore in project requirements [1]\n\n[1] https://opendev.org/openstack/adjutant/src/commit/df8c3e4a8b70be8697ac46f0acec9169752b4698/requirements.txt#L3\n\nChange-Id: Ie5c4f5528a25d0a13c12749c1d7af09377792342\n","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/openstack-ansible-os_adjutant/commit/ebca0060c12c6ecdb0f4052ff3f6022d01663864"}],"resolve_conflicts_web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/openstack-ansible-os_adjutant/commit/ebca0060c12c6ecdb0f4052ff3f6022d01663864"}]},"branch":"refs/heads/master","description":"Edit commit message"},"7444c804a9de8b71d568f63c35f47a5dda697267":{"kind":"REWORK","_number":7,"created":"2023-04-10 16:54:32.000000000","uploader":{"_account_id":28619,"name":"Dmitriy Rabotyagov","email":"noonedeadpunk@gmail.com","username":"noonedeadpunk"},"ref":"refs/changes/92/879892/7","fetch":{"anonymous http":{"url":"https://review.opendev.org/openstack/openstack-ansible-os_adjutant","ref":"refs/changes/92/879892/7","commands":{"Checkout":"git fetch https://review.opendev.org/openstack/openstack-ansible-os_adjutant refs/changes/92/879892/7 \u0026\u0026 git checkout FETCH_HEAD","Cherry Pick":"git fetch https://review.opendev.org/openstack/openstack-ansible-os_adjutant refs/changes/92/879892/7 \u0026\u0026 git cherry-pick FETCH_HEAD","Format Patch":"git fetch https://review.opendev.org/openstack/openstack-ansible-os_adjutant refs/changes/92/879892/7 \u0026\u0026 git format-patch -1 --stdout FETCH_HEAD","Pull":"git pull https://review.opendev.org/openstack/openstack-ansible-os_adjutant refs/changes/92/879892/7"}}},"commit":{"parents":[{"commit":"7f2d4a20736b63116c8ee03e10f0e2370544195c","subject":"Update master for stable/zed","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/openstack-ansible-os_adjutant/commit/7f2d4a20736b63116c8ee03e10f0e2370544195c"}]}],"author":{"name":"Dmitriy Rabotyagov","email":"dmitriy.rabotyagov@cleura.com","date":"2023-04-07 12:21:53.000000000","tz":120},"committer":{"name":"Dmitriy Rabotyagov","email":"noonedeadpunk@gmail.com","date":"2023-04-10 16:54:32.000000000","tz":0},"subject":"Switch back to use constrained Django version","message":"Switch back to use constrained Django version\n\nSupposedly, Adjutant has upgraded used Django version and constrained\nversion can be used now, since upper limit for Django is not defined\nanymore in project requirements [1]\n\nWe also fix ansible_facts usage with same commit to resolve\ncircular dependency.\n\n[1] https://opendev.org/openstack/adjutant/src/commit/df8c3e4a8b70be8697ac46f0acec9169752b4698/requirements.txt#L3\n\nChange-Id: Ie5c4f5528a25d0a13c12749c1d7af09377792342\n","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/openstack-ansible-os_adjutant/commit/7444c804a9de8b71d568f63c35f47a5dda697267"}],"resolve_conflicts_web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/openstack-ansible-os_adjutant/commit/7444c804a9de8b71d568f63c35f47a5dda697267"}]},"branch":"refs/heads/master"}},"requirements":[],"submit_records":[{"rule_name":"gerrit~DefaultSubmitRule","status":"CLOSED","labels":[{"label":"Verified","status":"MAY","applied_by":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]}},{"label":"Code-Review","status":"MAY","applied_by":{"_account_id":25023,"name":"Jonathan Rosser","email":"jonathan.rosser@rd.bbc.co.uk","username":"jrosser"}},{"label":"Workflow","status":"MAY","applied_by":{"_account_id":25023,"name":"Jonathan Rosser","email":"jonathan.rosser@rd.bbc.co.uk","username":"jrosser"}},{"label":"Backport-Candidate","status":"MAY"}]}],"submit_requirements":[{"name":"Verified","description":"Verified in gate by CI","status":"SATISFIED","is_legacy":false,"submittability_expression_result":{"expression":"label:Verified\u003dMAX AND -label:Verified\u003dMIN","fulfilled":true,"status":"PASS","passing_atoms":["label:Verified\u003dMAX"],"failing_atoms":["label:Verified\u003dMIN"],"atom_explanations":{}}},{"name":"Backport-Candidate","description":"Backport candidate status","status":"NOT_APPLICABLE","is_legacy":false,"applicability_expression_result":{"fulfilled":false,"status":"FAIL"},"submittability_expression_result":{"expression":"is:true","fulfilled":true,"status":"NOT_EVALUATED","passing_atoms":[],"failing_atoms":[],"atom_explanations":{}}},{"name":"Code-Review","description":"Code reviewed by core reviewer","status":"SATISFIED","is_legacy":false,"submittability_expression_result":{"expression":"label:Code-Review\u003dMAX AND -label:Code-Review\u003dMIN","fulfilled":true,"status":"PASS","passing_atoms":["label:Code-Review\u003dMAX"],"failing_atoms":["label:Code-Review\u003dMIN"],"atom_explanations":{}}},{"name":"Workflow","description":"Approved for gate by core reviewer","status":"SATISFIED","is_legacy":false,"submittability_expression_result":{"expression":"label:Workflow\u003dMAX AND -label:Workflow\u003dMIN","fulfilled":true,"status":"PASS","passing_atoms":["label:Workflow\u003dMAX"],"failing_atoms":["label:Workflow\u003dMIN"],"atom_explanations":{}}}]}
