)]}'
{"id":"openstack%2Fopenstack-ansible-os_adjutant~919679","triplet_id":"openstack%2Fopenstack-ansible-os_adjutant~master~I4b12bd4054e36213e94d7bcb53d4a5aceb439e42","project":"openstack/openstack-ansible-os_adjutant","branch":"master","topic":"osa/rmq-migrate","attention_set":{},"removed_from_attention_set":{"31542":{"account":{"_account_id":31542,"name":"Andrew Bonney","email":"andrew.bonney@bbc.co.uk","username":"andrewbonney"},"last_update":"2025-08-18 12:33:41.000000000","reason":"Change was submitted"},"32666":{"account":{"_account_id":32666,"name":"Damian Dąbrowski","email":"damian@dabrowski.cloud","username":"ddabrowski"},"last_update":"2025-08-18 12:33:41.000000000","reason":"Change was submitted"},"34411":{"account":{"_account_id":34411,"name":"Neil Hanlon","email":"neil@shrug.pw","username":"nhanlon"},"last_update":"2025-08-18 12:33:41.000000000","reason":"Change was submitted"},"25023":{"account":{"_account_id":25023,"name":"Jonathan Rosser","email":"jonathan.rosser@rd.bbc.co.uk","username":"jrosser"},"last_update":"2025-08-18 10:44:04.000000000","reason":"\u003cGERRIT_ACCOUNT_25023\u003e replied on the change","reason_account":{"_account_id":25023,"name":"Jonathan Rosser","email":"jonathan.rosser@rd.bbc.co.uk","username":"jrosser"}}},"hashtags":[],"change_id":"I4b12bd4054e36213e94d7bcb53d4a5aceb439e42","subject":"Add tag to enable targeting of post-install config elements only","status":"MERGED","created":"2024-05-15 08:16:46.000000000","updated":"2025-08-18 12:34:31.000000000","submitted":"2025-08-18 12:33:41.000000000","submitter":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"total_comment_count":2,"unresolved_comment_count":0,"has_review_started":true,"submission_id":"919679-osa/rmq-migrate","meta_rev_id":"244584977d8de116fe99399cd4a6ba49ceca8592","_number":919679,"virtual_id_number":919679,"owner":{"_account_id":31542,"name":"Andrew Bonney","email":"andrew.bonney@bbc.co.uk","username":"andrewbonney"},"actions":{},"labels":{"Verified":{"approved":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"all":[{"value":0,"_account_id":28619,"name":"Dmitriy Rabotyagov","email":"noonedeadpunk@gmail.com","username":"noonedeadpunk"},{"tag":"autogenerated:zuul:gate","value":2,"date":"2025-08-18 12:33:41.000000000","permitted_voting_range":{"min":2,"max":2},"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},{"value":0,"_account_id":37598,"name":"Ivan Anfimov","display_name":"Ivan Anfimov","email":"lazekteam@gmail.com","username":"anfimovir"},{"value":0,"_account_id":25023,"name":"Jonathan Rosser","email":"jonathan.rosser@rd.bbc.co.uk","username":"jrosser"}],"values":{"-2":"Fails","-1":"Doesn\u0027t seem to work"," 0":"No score","+1":"Works for me","+2":"Verified"},"description":"","default_value":0,"optional":true},"Code-Review":{"approved":{"_account_id":28619,"name":"Dmitriy Rabotyagov","email":"noonedeadpunk@gmail.com","username":"noonedeadpunk"},"all":[{"value":2,"date":"2025-08-15 06:38:11.000000000","permitted_voting_range":{"min":2,"max":2},"_account_id":28619,"name":"Dmitriy Rabotyagov","email":"noonedeadpunk@gmail.com","username":"noonedeadpunk"},{"value":0,"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},{"value":1,"date":"2025-08-15 10:55:10.000000000","permitted_voting_range":{"min":1,"max":2},"_account_id":37598,"name":"Ivan Anfimov","display_name":"Ivan Anfimov","email":"lazekteam@gmail.com","username":"anfimovir"},{"value":2,"date":"2025-08-18 10:44:04.000000000","permitted_voting_range":{"min":2,"max":2},"_account_id":25023,"name":"Jonathan Rosser","email":"jonathan.rosser@rd.bbc.co.uk","username":"jrosser"}],"values":{"-2":"Do not merge","-1":"This patch needs further work before it can be merged"," 0":"No score","+1":"Looks good to me, but someone else must approve","+2":"Looks good to me (core reviewer)"},"description":"","default_value":0,"optional":true},"Workflow":{"approved":{"_account_id":25023,"name":"Jonathan Rosser","email":"jonathan.rosser@rd.bbc.co.uk","username":"jrosser"},"all":[{"value":0,"_account_id":28619,"name":"Dmitriy Rabotyagov","email":"noonedeadpunk@gmail.com","username":"noonedeadpunk"},{"value":0,"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},{"value":0,"_account_id":37598,"name":"Ivan Anfimov","display_name":"Ivan Anfimov","email":"lazekteam@gmail.com","username":"anfimovir"},{"value":1,"date":"2025-08-18 10:44:04.000000000","permitted_voting_range":{"min":1,"max":1},"_account_id":25023,"name":"Jonathan Rosser","email":"jonathan.rosser@rd.bbc.co.uk","username":"jrosser"}],"values":{"-1":"Work in progress"," 0":"Ready for reviews","+1":"Approved"},"description":"","default_value":0,"optional":true},"Backport-Candidate":{"all":[{"value":0,"_account_id":28619,"name":"Dmitriy Rabotyagov","email":"noonedeadpunk@gmail.com","username":"noonedeadpunk"},{"value":0,"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},{"value":0,"_account_id":37598,"name":"Ivan Anfimov","display_name":"Ivan Anfimov","email":"lazekteam@gmail.com","username":"anfimovir"},{"value":0,"_account_id":25023,"name":"Jonathan Rosser","email":"jonathan.rosser@rd.bbc.co.uk","username":"jrosser"}],"values":{"-1":"Not A Backport Candidate"," 0":"Backport Review Needed","+1":"Proposed Backport"},"description":"","default_value":0,"optional":true}},"removable_reviewers":[{"_account_id":32666,"name":"Damian Dąbrowski","email":"damian@dabrowski.cloud","username":"ddabrowski"},{"_account_id":34411,"name":"Neil Hanlon","email":"neil@shrug.pw","username":"nhanlon"}],"reviewers":{"REVIEWER":[{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},{"_account_id":25023,"name":"Jonathan Rosser","email":"jonathan.rosser@rd.bbc.co.uk","username":"jrosser"},{"_account_id":28619,"name":"Dmitriy Rabotyagov","email":"noonedeadpunk@gmail.com","username":"noonedeadpunk"},{"_account_id":32666,"name":"Damian Dąbrowski","email":"damian@dabrowski.cloud","username":"ddabrowski"},{"_account_id":34411,"name":"Neil Hanlon","email":"neil@shrug.pw","username":"nhanlon"},{"_account_id":37598,"name":"Ivan Anfimov","display_name":"Ivan Anfimov","email":"lazekteam@gmail.com","username":"anfimovir"}]},"pending_reviewers":{},"reviewer_updates":[{"updated":"2024-05-15 11:04:01.000000000","updated_by":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"reviewer":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"state":"REVIEWER"},{"updated":"2025-08-15 06:38:11.000000000","updated_by":{"_account_id":28619,"name":"Dmitriy Rabotyagov","email":"noonedeadpunk@gmail.com","username":"noonedeadpunk"},"reviewer":{"_account_id":28619,"name":"Dmitriy Rabotyagov","email":"noonedeadpunk@gmail.com","username":"noonedeadpunk"},"state":"REVIEWER"},{"updated":"2025-08-15 09:03:41.000000000","updated_by":{"_account_id":37598,"name":"Ivan Anfimov","display_name":"Ivan Anfimov","email":"lazekteam@gmail.com","username":"anfimovir"},"reviewer":{"_account_id":37598,"name":"Ivan Anfimov","display_name":"Ivan Anfimov","email":"lazekteam@gmail.com","username":"anfimovir"},"state":"REVIEWER"},{"updated":"2025-08-15 10:55:01.000000000","updated_by":{"_account_id":37598,"name":"Ivan Anfimov","display_name":"Ivan Anfimov","email":"lazekteam@gmail.com","username":"anfimovir"},"reviewer":{"_account_id":37598,"name":"Ivan Anfimov","display_name":"Ivan Anfimov","email":"lazekteam@gmail.com","username":"anfimovir"},"state":"REMOVED"},{"updated":"2025-08-15 10:55:01.000000000","updated_by":{"_account_id":37598,"name":"Ivan Anfimov","display_name":"Ivan Anfimov","email":"lazekteam@gmail.com","username":"anfimovir"},"reviewer":{"_account_id":34411,"name":"Neil Hanlon","email":"neil@shrug.pw","username":"nhanlon"},"state":"REVIEWER"},{"updated":"2025-08-15 10:55:01.000000000","updated_by":{"_account_id":37598,"name":"Ivan Anfimov","display_name":"Ivan Anfimov","email":"lazekteam@gmail.com","username":"anfimovir"},"reviewer":{"_account_id":32666,"name":"Damian Dąbrowski","email":"damian@dabrowski.cloud","username":"ddabrowski"},"state":"REVIEWER"},{"updated":"2025-08-15 10:55:01.000000000","updated_by":{"_account_id":37598,"name":"Ivan Anfimov","display_name":"Ivan Anfimov","email":"lazekteam@gmail.com","username":"anfimovir"},"reviewer":{"_account_id":25023,"name":"Jonathan Rosser","email":"jonathan.rosser@rd.bbc.co.uk","username":"jrosser"},"state":"REVIEWER"},{"updated":"2025-08-15 10:55:10.000000000","updated_by":{"_account_id":37598,"name":"Ivan Anfimov","display_name":"Ivan Anfimov","email":"lazekteam@gmail.com","username":"anfimovir"},"reviewer":{"_account_id":37598,"name":"Ivan Anfimov","display_name":"Ivan Anfimov","email":"lazekteam@gmail.com","username":"anfimovir"},"state":"REVIEWER"}],"messages":[{"id":"cbc8a836637c917bf9b05eb8c1d1a3bc34afbae5","tag":"autogenerated:gerrit:newPatchSet","author":{"_account_id":31542,"name":"Andrew Bonney","email":"andrew.bonney@bbc.co.uk","username":"andrewbonney"},"date":"2024-05-15 08:16:46.000000000","message":"Uploaded patch set 1.","accounts_in_message":[],"_revision_number":1},{"id":"d3e748fc8218e6a5ce1508dd21aed1cbb4ff14ab","tag":"autogenerated:zuul:check","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2024-05-15 11:04:01.000000000","message":"Patch Set 1: Verified-1\n\nBuild failed (check pipeline).  For information on how to proceed, see\nhttps://docs.opendev.org/opendev/infra-manual/latest/developers.html#automated-testing\nand https://docs.openstack.org/project-team-guide/testing.html#how-to-handle-test-failures\n\nhttps://zuul.opendev.org/t/openstack/buildset/8c3d3cb6f9234a67bcec3330e1b69cea\n\n- openstack-ansible-linters-ubuntu-jammy https://zuul.opendev.org/t/openstack/build/1c0dd77b2f304b209c17ed86c01cf534 : SUCCESS in 18m 00s\n- openstack-ansible-deploy-aio_metal-debian-bullseye https://zuul.opendev.org/t/openstack/build/13a79c2a4c7349cf96231b14c3a48cc7 : FAILURE in 1h 08m 52s\n- openstack-ansible-deploy-aio_metal-debian-bookworm https://zuul.opendev.org/t/openstack/build/dae8599269c746768827998b1129e7b1 : FAILURE in 1h 11m 19s\n- openstack-ansible-deploy-aio_metal-centos-9-stream https://zuul.opendev.org/t/openstack/build/bc9e45dc637f4949a63501f644d41c3d : FAILURE in 1h 10m 32s (non-voting)\n- openstack-ansible-deploy-aio_metal-rockylinux-9 https://zuul.opendev.org/t/openstack/build/6d85e37648954f4982b092612a764e08 : FAILURE in 1h 10m 59s\n- openstack-ansible-deploy-aio_metal-ubuntu-jammy https://zuul.opendev.org/t/openstack/build/dd4ac46491ce41e086af1a02cf99a0b1 : FAILURE in 1h 09m 21s\n- openstack-ansible-deploy-aio_metal_tls-rockylinux-9 https://zuul.opendev.org/t/openstack/build/6fe428974c1f423f9c6165ad5b20b9a5 : FAILURE in 1h 12m 49s\n- openstack-ansible-deploy-aio_metal_tls-ubuntu-jammy https://zuul.opendev.org/t/openstack/build/528a96fa533e420182f419bdfb37d0b9 : FAILURE in 1h 06m 52s\n- openstack-ansible-deploy-aio_metal_quorum-ubuntu-jammy https://zuul.opendev.org/t/openstack/build/506466aaac0445da98bda7f6bfd8dbfe : FAILURE in 1h 01m 07s\n- openstack-ansible-upgrade-aio_metal-rockylinux-9 https://zuul.opendev.org/t/openstack/build/a228e44cf2784126a5c18b566ca37a1c : FAILURE in 1h 50m 55s\n- openstack-ansible-upgrade_2023.1-aio_metal-rockylinux-9 https://zuul.opendev.org/t/openstack/build/6d6c81af111648cf94ce5b2b00894b72 : FAILURE in 2h 12m 01s\n- openstack-ansible-upgrade-aio_metal-ubuntu-jammy https://zuul.opendev.org/t/openstack/build/f5673ec3de8141b69eb8265cf6663585 : FAILURE in 1h 36m 23s\n- openstack-ansible-upgrade_2023.1-aio_metal-ubuntu-jammy https://zuul.opendev.org/t/openstack/build/10482a88e832483eb77bf2c4d8ea0838 : FAILURE in 2h 22m 07s\n- openstack-ansible-upgrade-aio_metal_tls-rockylinux-9 https://zuul.opendev.org/t/openstack/build/7ea1ac6fee80444d9867a1de5078acab : FAILURE in 2h 08m 56s\n- openstack-ansible-upgrade-aio_metal_tls-ubuntu-jammy https://zuul.opendev.org/t/openstack/build/b5d26b888b0c4f63a254ef781b894b7b : FAILURE in 1h 45m 29s\n- openstack-tox-docs https://zuul.opendev.org/t/openstack/build/c5f617f0e4ca4867a316a9467452dc41 : SUCCESS in 4m 40s","accounts_in_message":[],"_revision_number":1},{"id":"1c3b7d4cace507d0dab80772b64fac52fc861e70","tag":"autogenerated:gerrit:newPatchSet","author":{"_account_id":31542,"name":"Andrew Bonney","email":"andrew.bonney@bbc.co.uk","username":"andrewbonney"},"real_author":{"_account_id":37598,"name":"Ivan Anfimov","display_name":"Ivan Anfimov","email":"lazekteam@gmail.com","username":"anfimovir"},"date":"2025-06-02 15:31:03.000000000","message":"Patch Set 2: Patch Set 1 was rebased on behalf of \u003cGERRIT_ACCOUNT_31542\u003e\n\nOutdated Votes:\n* Verified-1\n","accounts_in_message":[{"_account_id":31542,"name":"Andrew Bonney","email":"andrew.bonney@bbc.co.uk","username":"andrewbonney"}],"_revision_number":2},{"id":"382b1bd775ad54237ba4d035ddf089ab193fd88b","tag":"autogenerated:zuul:check","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2025-06-02 17:33:17.000000000","message":"Patch Set 2: Verified-1\n\nBuild failed (check pipeline).  For information on how to proceed, see\nhttps://docs.opendev.org/opendev/infra-manual/latest/developers.html#automated-testing\nand https://docs.openstack.org/project-team-guide/testing.html#how-to-handle-test-failures\n\nhttps://zuul.opendev.org/t/openstack/buildset/f12ee64bffa7460fa1030280aae03de1\n\n- openstack-ansible-linters-ubuntu-noble https://zuul.opendev.org/t/openstack/build/2883c8fe808a48fea95ad28f2224b287 : SUCCESS in 7m 51s\n- openstack-ansible-deploy-aio_metal-debian-bookworm https://zuul.opendev.org/t/openstack/build/ef5d5f9f7195420f98361df1e503b5b6 : SUCCESS in 1h 14m 30s\n- openstack-ansible-deploy-aio_metal-ubuntu-noble https://zuul.opendev.org/t/openstack/build/de4154ed2dd4455c9d8486e7ece8f2be : SUCCESS in 1h 17m 29s\n- openstack-ansible-deploy-aio_metal_tls-ubuntu-noble https://zuul.opendev.org/t/openstack/build/da8d090403bd469b940bba04bb9e3678 : FAILURE in 1h 08m 07s\n- openstack-ansible-upgrade-aio_metal-ubuntu-noble https://zuul.opendev.org/t/openstack/build/d3aa4ae5077946829379fbf36298555d : SUCCESS in 1h 58m 45s\n- openstack-ansible-upgrade-aio_metal_tls-ubuntu-noble https://zuul.opendev.org/t/openstack/build/9e41a57b0f8a474e84e55cad62e8fc60 : FAILURE in 1h 59m 10s\n- openstack-tox-docs https://zuul.opendev.org/t/openstack/build/5f5bc97fbcfb45b3aed3ec5bff91aa50 : SUCCESS in 5m 55s","accounts_in_message":[],"_revision_number":2},{"id":"2f121048a400fe99413e89442faacbc1f07d7423","tag":"autogenerated:gerrit:newPatchSet","author":{"_account_id":31542,"name":"Andrew Bonney","email":"andrew.bonney@bbc.co.uk","username":"andrewbonney"},"date":"2025-08-13 14:17:36.000000000","message":"Uploaded patch set 3.\n\nOutdated Votes:\n* Verified-1\n","accounts_in_message":[],"_revision_number":3},{"id":"a228b2a25b7afea2ad6fe7f8cbcd884e442dea2d","tag":"autogenerated:zuul:check","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2025-08-13 19:55:34.000000000","message":"Patch Set 3: Verified-1\n\nBuild failed (check pipeline).  For information on how to proceed, see\nhttps://docs.opendev.org/opendev/infra-manual/latest/developers.html#automated-testing\nand https://docs.openstack.org/project-team-guide/testing.html#how-to-handle-test-failures\n\nhttps://zuul.opendev.org/t/openstack/buildset/54c9a626d32b4255b941e133408a5a79\n\n- openstack-ansible-linters-ubuntu-noble https://zuul.opendev.org/t/openstack/build/5916817a9d614f2ca7bbebcbd5b403dd : SUCCESS in 4m 11s\n- openstack-ansible-deploy-aio_metal-debian-bookworm https://zuul.opendev.org/t/openstack/build/c0a756be966e4471b5a68bbd7d0fe4ba : SUCCESS in 41m 42s\n- openstack-ansible-deploy-aio_metal-ubuntu-noble https://zuul.opendev.org/t/openstack/build/86f690bd19824aa38a4f7b989fcb3fe1 : SUCCESS in 44m 11s\n- openstack-ansible-deploy-aio_metal_tls-ubuntu-noble https://zuul.opendev.org/t/openstack/build/87f3deeb396a4e52934f765852a6fd5d : FAILURE in 57m 57s\n- openstack-ansible-upgrade-aio_metal-ubuntu-noble https://zuul.opendev.org/t/openstack/build/7d39278befbc4853b574cae5866e0601 : SUCCESS in 56m 21s\n- openstack-ansible-upgrade-aio_metal_tls-ubuntu-noble https://zuul.opendev.org/t/openstack/build/af7bc51bdd8d4f3190cdbfa4d622bb1a : FAILURE in 27m 59s\n- openstack-tox-docs https://zuul.opendev.org/t/openstack/build/034a5f4ddb274eb8a1f0410ced7c3ab8 : SUCCESS in 3m 13s","accounts_in_message":[],"_revision_number":3},{"id":"f0252b5aea8d217fc4cfd899364156e53d2c14c2","tag":"autogenerated:gerrit:newPatchSet","author":{"_account_id":31542,"name":"Andrew Bonney","email":"andrew.bonney@bbc.co.uk","username":"andrewbonney"},"real_author":{"_account_id":28619,"name":"Dmitriy Rabotyagov","email":"noonedeadpunk@gmail.com","username":"noonedeadpunk"},"date":"2025-08-15 06:38:04.000000000","message":"Patch Set 4: Patch Set 3 was rebased on behalf of \u003cGERRIT_ACCOUNT_31542\u003e\n\nOutdated Votes:\n* Verified-1\n","accounts_in_message":[{"_account_id":31542,"name":"Andrew Bonney","email":"andrew.bonney@bbc.co.uk","username":"andrewbonney"}],"_revision_number":4},{"id":"cf27875e4a63f86b90b32cae6de96fa43ff5e3d0","author":{"_account_id":28619,"name":"Dmitriy Rabotyagov","email":"noonedeadpunk@gmail.com","username":"noonedeadpunk"},"date":"2025-08-15 06:38:11.000000000","message":"Patch Set 4: Code-Review+2","accounts_in_message":[],"_revision_number":4},{"id":"ce489076fd7ae92ad2c26a8749154dfbd39361b1","author":{"_account_id":37598,"name":"Ivan Anfimov","display_name":"Ivan Anfimov","email":"lazekteam@gmail.com","username":"anfimovir"},"date":"2025-08-15 09:03:41.000000000","message":"Patch Set 4: Code-Review+1\n\n(1 comment)","accounts_in_message":[],"_revision_number":4},{"id":"ae712b2637d9240512b772628b83af77555e7a40","tag":"autogenerated:zuul:check","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2025-08-15 09:04:20.000000000","message":"Patch Set 4: Verified-1\n\nBuild failed (check pipeline).  For information on how to proceed, see\nhttps://docs.opendev.org/opendev/infra-manual/latest/developers.html#automated-testing\nand https://docs.openstack.org/project-team-guide/testing.html#how-to-handle-test-failures\n\nhttps://zuul.opendev.org/t/openstack/buildset/5c9b56adf5cb4922ac2e4b5981a5b886\n\n- openstack-ansible-linters-ubuntu-noble https://zuul.opendev.org/t/openstack/build/aa2bc148cc8f41b4bfeefdf94f602e49 : SUCCESS in 5m 12s\n- openstack-ansible-deploy-aio_metal-debian-bookworm https://zuul.opendev.org/t/openstack/build/1c84103654a7437bae0366f3945aeb79 : FAILURE in 1h 21m 38s\n- openstack-ansible-deploy-aio_metal-ubuntu-noble https://zuul.opendev.org/t/openstack/build/1b4dc9984bdd45d4b4e47ca29b8196b8 : SUCCESS in 49m 27s\n- openstack-ansible-deploy-aio_metal_tls-ubuntu-noble https://zuul.opendev.org/t/openstack/build/e668e83884dc45a7ab874dc3712e4b36 : SUCCESS in 46m 52s\n- openstack-ansible-upgrade-aio_metal-ubuntu-noble https://zuul.opendev.org/t/openstack/build/38d527b5172b454dacd6a65f6d185666 : SUCCESS in 1h 12m 04s\n- openstack-ansible-upgrade-aio_metal_tls-ubuntu-noble https://zuul.opendev.org/t/openstack/build/1810241884864cb3ae0bddc63a6ebc4d : SUCCESS in 2h 22m 51s\n- openstack-tox-docs https://zuul.opendev.org/t/openstack/build/bccd4f26f08942a98612e1ae1163f156 : SUCCESS in 5m 47s","accounts_in_message":[],"_revision_number":4},{"id":"cecbbeb0dd0aa67ce2725a175ad2592c2a18a90b","author":{"_account_id":37598,"name":"Ivan Anfimov","display_name":"Ivan Anfimov","email":"lazekteam@gmail.com","username":"anfimovir"},"date":"2025-08-15 09:04:52.000000000","message":"Patch Set 4:\n\n(1 comment)","accounts_in_message":[],"_revision_number":4},{"id":"3a35830d76f726f946032eed8f659a42280b6f3b","tag":"autogenerated:zuul:check","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2025-08-15 10:54:32.000000000","message":"Patch Set 4: Verified+1\n\nBuild succeeded (check pipeline).\nhttps://zuul.opendev.org/t/openstack/buildset/3a9cd98fb1f043aab4665dc6c0cc5646\n\n- openstack-ansible-linters-ubuntu-noble https://zuul.opendev.org/t/openstack/build/ec18c1587cb44b4e943d569caf3fd8cf : SUCCESS in 6m 17s\n- openstack-ansible-deploy-aio_metal-debian-bookworm https://zuul.opendev.org/t/openstack/build/8f4a002c1e794b188c912220eb6fdaab : SUCCESS in 40m 04s\n- openstack-ansible-deploy-aio_metal-ubuntu-noble https://zuul.opendev.org/t/openstack/build/0ac0d31da9354129a99216df75fe004e : SUCCESS in 44m 15s\n- openstack-ansible-deploy-aio_metal_tls-ubuntu-noble https://zuul.opendev.org/t/openstack/build/ebc7bc27c88c48f0b1ee28cf7161f8c9 : SUCCESS in 1h 44m 38s\n- openstack-ansible-upgrade-aio_metal-ubuntu-noble https://zuul.opendev.org/t/openstack/build/707ca4767eec48ad92b961b757b225dc : SUCCESS in 1h 47m 12s\n- openstack-ansible-upgrade-aio_metal_tls-ubuntu-noble https://zuul.opendev.org/t/openstack/build/695f3ba381b941efa02d2617ea62bdcf : SUCCESS in 1h 15m 26s\n- openstack-tox-docs https://zuul.opendev.org/t/openstack/build/37abc6e7dfe641afa7215c56db76c09f : SUCCESS in 4m 32s","accounts_in_message":[],"_revision_number":4},{"id":"f32418f83a2f71749ada356f4c974152519790c6","tag":"autogenerated:gerrit:deleteReviewer","author":{"_account_id":37598,"name":"Ivan Anfimov","display_name":"Ivan Anfimov","email":"lazekteam@gmail.com","username":"anfimovir"},"date":"2025-08-15 10:55:01.000000000","message":"Removed reviewer \u003cGERRIT_ACCOUNT_37598\u003e with the following votes:\n\n* Code-Review+1 by \u003cGERRIT_ACCOUNT_37598\u003e\n","accounts_in_message":[{"_account_id":37598,"name":"Ivan Anfimov","display_name":"Ivan Anfimov","email":"lazekteam@gmail.com","username":"anfimovir"}],"_revision_number":4},{"id":"eec35f6e50f85ca1f3983cc10d7173607fbd4079","author":{"_account_id":37598,"name":"Ivan Anfimov","display_name":"Ivan Anfimov","email":"lazekteam@gmail.com","username":"anfimovir"},"date":"2025-08-15 10:55:10.000000000","message":"Patch Set 4: Code-Review+1","accounts_in_message":[],"_revision_number":4},{"id":"57d738977066b2d7eeabc01e59a3619ed3e90d69","author":{"_account_id":25023,"name":"Jonathan Rosser","email":"jonathan.rosser@rd.bbc.co.uk","username":"jrosser"},"date":"2025-08-18 10:44:04.000000000","message":"Patch Set 4: Code-Review+2 Workflow+1","accounts_in_message":[],"_revision_number":4},{"id":"3f2453d0beda40c4bd03790a1dc8e8a43d3f33d7","tag":"autogenerated:zuul:gate","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2025-08-18 10:46:03.000000000","message":"Patch Set 4: -Verified\n\nStarting gate jobs.","accounts_in_message":[],"_revision_number":4},{"id":"ebf7247cb475ddc312d4a9d1cd52cd1d16bef3db","tag":"autogenerated:zuul:gate","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2025-08-18 12:33:41.000000000","message":"Patch Set 4: Verified+2\n\nBuild succeeded (gate pipeline).\nhttps://zuul.opendev.org/t/openstack/buildset/b7dac48e984e46a0911d589e1f6837bb\n\n- openstack-ansible-linters-ubuntu-noble https://zuul.opendev.org/t/openstack/build/2d78b015ba6141e490cb26144c949f6c : SUCCESS in 9m 44s\n- openstack-ansible-deploy-aio_metal-centos-10-stream https://zuul.opendev.org/t/openstack/build/5447a225abc04656b2d1057afa8ca37d : SUCCESS in 1h 17m 11s\n- openstack-ansible-deploy-aio_metal-debian-bookworm https://zuul.opendev.org/t/openstack/build/c8b6a0b826cf44c68deb3d527484fc2d : SUCCESS in 42m 25s\n- openstack-ansible-deploy-aio_metal-ubuntu-noble https://zuul.opendev.org/t/openstack/build/9df8f13cbbf34acea45512e775bdda15 : SUCCESS in 42m 11s\n- openstack-ansible-deploy-aio_metal_tls-ubuntu-noble https://zuul.opendev.org/t/openstack/build/711bce20b0ad44bdb264f5e725ce4233 : SUCCESS in 1h 41m 34s\n- openstack-tox-docs https://zuul.opendev.org/t/openstack/build/effce61b44d64ad892bacc7147bb55d4 : SUCCESS in 6m 40s","accounts_in_message":[],"_revision_number":4},{"id":"0c66e516aea0fdf2266cf6b8deb888904a94a7b0","tag":"autogenerated:gerrit:merged","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2025-08-18 12:33:41.000000000","message":"Change has been successfully merged","accounts_in_message":[],"_revision_number":4},{"id":"244584977d8de116fe99399cd4a6ba49ceca8592","tag":"autogenerated:zuul:promote","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2025-08-18 12:34:31.000000000","message":"Patch Set 4:\n\nBuild succeeded (promote pipeline).\nhttps://zuul.opendev.org/t/openstack/buildset/ee7bfb2b1664438bb30ac7274c7115fa\n\n- promote-openstack-tox-docs https://zuul.opendev.org/t/openstack/build/9a4f8761291e4cc1836bb99745985098 : SUCCESS in 41s","accounts_in_message":[],"_revision_number":4}],"current_revision_number":4,"current_revision":"940cd2dd0e0e2e2ab96e5ae9ea95c7c1cb2987f5","revisions":{"ebdfec6a05f3ced26e931b76a40e6b4180996164":{"kind":"REWORK","_number":1,"created":"2024-05-15 08:16:46.000000000","uploader":{"_account_id":31542,"name":"Andrew Bonney","email":"andrew.bonney@bbc.co.uk","username":"andrewbonney"},"ref":"refs/changes/79/919679/1","fetch":{"anonymous http":{"url":"https://review.opendev.org/openstack/openstack-ansible-os_adjutant","ref":"refs/changes/79/919679/1","commands":{"Checkout":"git fetch https://review.opendev.org/openstack/openstack-ansible-os_adjutant refs/changes/79/919679/1 \u0026\u0026 git checkout FETCH_HEAD","Cherry Pick":"git fetch https://review.opendev.org/openstack/openstack-ansible-os_adjutant refs/changes/79/919679/1 \u0026\u0026 git cherry-pick FETCH_HEAD","Format Patch":"git fetch https://review.opendev.org/openstack/openstack-ansible-os_adjutant refs/changes/79/919679/1 \u0026\u0026 git format-patch -1 --stdout FETCH_HEAD","Pull":"git pull https://review.opendev.org/openstack/openstack-ansible-os_adjutant refs/changes/79/919679/1"}}},"commit":{"parents":[{"commit":"81c081eb18bd48d9e2d1be803e118b1b26c0c005","subject":"Merge \"Replace deprecated UPPER_CONSTRAINTS_FILE variable\"","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/openstack-ansible-os_adjutant/commit/81c081eb18bd48d9e2d1be803e118b1b26c0c005"}]}],"author":{"name":"Andrew Bonney","email":"andrew.bonney@bbc.co.uk","date":"2024-05-15 08:09:17.000000000","tz":60},"committer":{"name":"Andrew Bonney","email":"andrew.bonney@bbc.co.uk","date":"2024-05-15 08:16:44.000000000","tz":60},"subject":"Add tag to enable targeting of post-install config elements only","message":"Add tag to enable targeting of post-install config elements only\n\n\u003cservice\u003e-config tags are quite broad and have a long execution\ntime. Where you only need to modify a service\u0027s \u0027.conf\u0027 file and\nsimilar it is useful to have a quicker method to do so.\n\nChange-Id: I4b12bd4054e36213e94d7bcb53d4a5aceb439e42\n","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/openstack-ansible-os_adjutant/commit/ebdfec6a05f3ced26e931b76a40e6b4180996164"}],"resolve_conflicts_web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/openstack-ansible-os_adjutant/commit/ebdfec6a05f3ced26e931b76a40e6b4180996164"}]},"branch":"refs/heads/master"},"d53c3ae1a7745fcfdf17659bd57a3890caa90669":{"kind":"TRIVIAL_REBASE","_number":2,"created":"2025-06-02 15:31:03.000000000","uploader":{"_account_id":31542,"name":"Andrew Bonney","email":"andrew.bonney@bbc.co.uk","username":"andrewbonney"},"real_uploader":{"_account_id":37598,"name":"Ivan Anfimov","display_name":"Ivan Anfimov","email":"lazekteam@gmail.com","username":"anfimovir"},"ref":"refs/changes/79/919679/2","fetch":{"anonymous http":{"url":"https://review.opendev.org/openstack/openstack-ansible-os_adjutant","ref":"refs/changes/79/919679/2","commands":{"Checkout":"git fetch https://review.opendev.org/openstack/openstack-ansible-os_adjutant refs/changes/79/919679/2 \u0026\u0026 git checkout FETCH_HEAD","Cherry Pick":"git fetch https://review.opendev.org/openstack/openstack-ansible-os_adjutant refs/changes/79/919679/2 \u0026\u0026 git cherry-pick FETCH_HEAD","Format Patch":"git fetch https://review.opendev.org/openstack/openstack-ansible-os_adjutant refs/changes/79/919679/2 \u0026\u0026 git format-patch -1 --stdout FETCH_HEAD","Pull":"git pull https://review.opendev.org/openstack/openstack-ansible-os_adjutant refs/changes/79/919679/2"}}},"commit":{"parents":[{"commit":"bf3eca428c3a081b926ad9a1df4c8ea3d590bd2f","subject":"Fix for usage of modules via FQCN","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/openstack-ansible-os_adjutant/commit/bf3eca428c3a081b926ad9a1df4c8ea3d590bd2f"}]}],"author":{"name":"Andrew Bonney","email":"andrew.bonney@bbc.co.uk","date":"2024-05-15 08:09:17.000000000","tz":60},"committer":{"name":"Andrew Bonney","email":"andrew.bonney@bbc.co.uk","date":"2025-06-02 15:31:03.000000000","tz":0},"subject":"Add tag to enable targeting of post-install config elements only","message":"Add tag to enable targeting of post-install config elements only\n\n\u003cservice\u003e-config tags are quite broad and have a long execution\ntime. Where you only need to modify a service\u0027s \u0027.conf\u0027 file and\nsimilar it is useful to have a quicker method to do so.\n\nChange-Id: I4b12bd4054e36213e94d7bcb53d4a5aceb439e42\n","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/openstack-ansible-os_adjutant/commit/d53c3ae1a7745fcfdf17659bd57a3890caa90669"}],"resolve_conflicts_web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/openstack-ansible-os_adjutant/commit/d53c3ae1a7745fcfdf17659bd57a3890caa90669"}]},"branch":"refs/heads/master","description":"Rebase"},"954232e2f67455eef342b0de9ec3550450932fa6":{"kind":"TRIVIAL_REBASE_WITH_MESSAGE_UPDATE","_number":3,"created":"2025-08-13 14:17:36.000000000","uploader":{"_account_id":31542,"name":"Andrew Bonney","email":"andrew.bonney@bbc.co.uk","username":"andrewbonney"},"ref":"refs/changes/79/919679/3","fetch":{"anonymous http":{"url":"https://review.opendev.org/openstack/openstack-ansible-os_adjutant","ref":"refs/changes/79/919679/3","commands":{"Checkout":"git fetch https://review.opendev.org/openstack/openstack-ansible-os_adjutant refs/changes/79/919679/3 \u0026\u0026 git checkout FETCH_HEAD","Cherry Pick":"git fetch https://review.opendev.org/openstack/openstack-ansible-os_adjutant refs/changes/79/919679/3 \u0026\u0026 git cherry-pick FETCH_HEAD","Format Patch":"git fetch https://review.opendev.org/openstack/openstack-ansible-os_adjutant refs/changes/79/919679/3 \u0026\u0026 git format-patch -1 --stdout FETCH_HEAD","Pull":"git pull https://review.opendev.org/openstack/openstack-ansible-os_adjutant refs/changes/79/919679/3"}}},"commit":{"parents":[{"commit":"e89e3b34f1bb7f152c9cde5221f45dfdf503ef22","subject":"reno: Update master for unmaintained/zed","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/openstack-ansible-os_adjutant/commit/e89e3b34f1bb7f152c9cde5221f45dfdf503ef22"}]}],"author":{"name":"Andrew Bonney","email":"andrew.bonney@bbc.co.uk","date":"2024-05-15 08:09:17.000000000","tz":60},"committer":{"name":"Andrew Bonney","email":"andrew.bonney@bbc.co.uk","date":"2025-08-13 14:17:07.000000000","tz":60},"subject":"Add tag to enable targeting of post-install config elements only","message":"Add tag to enable targeting of post-install config elements only\n\n\u003cservice\u003e-config tags are quite broad and have a long execution\ntime. Where you only need to modify a service\u0027s \u0027.conf\u0027 file and\nsimilar it is useful to have a quicker method to do so.\n\nChange-Id: I4b12bd4054e36213e94d7bcb53d4a5aceb439e42\nDepends-On: https://review.opendev.org/c/openstack/openstack-ansible-os_adjutant/+/951529\nSigned-off-by: Andrew Bonney \u003candrew.bonney@bbc.co.uk\u003e\n","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/openstack-ansible-os_adjutant/commit/954232e2f67455eef342b0de9ec3550450932fa6"}],"resolve_conflicts_web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/openstack-ansible-os_adjutant/commit/954232e2f67455eef342b0de9ec3550450932fa6"}]},"branch":"refs/heads/master"},"940cd2dd0e0e2e2ab96e5ae9ea95c7c1cb2987f5":{"kind":"TRIVIAL_REBASE","_number":4,"created":"2025-08-15 06:38:04.000000000","uploader":{"_account_id":31542,"name":"Andrew Bonney","email":"andrew.bonney@bbc.co.uk","username":"andrewbonney"},"real_uploader":{"_account_id":28619,"name":"Dmitriy Rabotyagov","email":"noonedeadpunk@gmail.com","username":"noonedeadpunk"},"ref":"refs/changes/79/919679/4","fetch":{"anonymous http":{"url":"https://review.opendev.org/openstack/openstack-ansible-os_adjutant","ref":"refs/changes/79/919679/4","commands":{"Checkout":"git fetch https://review.opendev.org/openstack/openstack-ansible-os_adjutant refs/changes/79/919679/4 \u0026\u0026 git checkout FETCH_HEAD","Cherry Pick":"git fetch https://review.opendev.org/openstack/openstack-ansible-os_adjutant refs/changes/79/919679/4 \u0026\u0026 git cherry-pick FETCH_HEAD","Format Patch":"git fetch https://review.opendev.org/openstack/openstack-ansible-os_adjutant refs/changes/79/919679/4 \u0026\u0026 git format-patch -1 --stdout FETCH_HEAD","Pull":"git pull https://review.opendev.org/openstack/openstack-ansible-os_adjutant refs/changes/79/919679/4"}}},"commit":{"parents":[{"commit":"31264633ecf89cf0433d443846001c20e964e186","subject":"Ensure that first/last host detection is deterministic","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/openstack-ansible-os_adjutant/commit/31264633ecf89cf0433d443846001c20e964e186"}]}],"author":{"name":"Andrew Bonney","email":"andrew.bonney@bbc.co.uk","date":"2024-05-15 08:09:17.000000000","tz":60},"committer":{"name":"Andrew Bonney","email":"andrew.bonney@bbc.co.uk","date":"2025-08-15 06:38:04.000000000","tz":0},"subject":"Add tag to enable targeting of post-install config elements only","message":"Add tag to enable targeting of post-install config elements only\n\n\u003cservice\u003e-config tags are quite broad and have a long execution\ntime. Where you only need to modify a service\u0027s \u0027.conf\u0027 file and\nsimilar it is useful to have a quicker method to do so.\n\nChange-Id: I4b12bd4054e36213e94d7bcb53d4a5aceb439e42\nDepends-On: https://review.opendev.org/c/openstack/openstack-ansible-os_adjutant/+/951529\nSigned-off-by: Andrew Bonney \u003candrew.bonney@bbc.co.uk\u003e\n","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/openstack-ansible-os_adjutant/commit/940cd2dd0e0e2e2ab96e5ae9ea95c7c1cb2987f5"}],"resolve_conflicts_web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/openstack-ansible-os_adjutant/commit/940cd2dd0e0e2e2ab96e5ae9ea95c7c1cb2987f5"}]},"branch":"refs/heads/master","description":"Rebase"}},"requirements":[],"submit_records":[{"rule_name":"gerrit~DefaultSubmitRule","status":"CLOSED","labels":[{"label":"Verified","status":"MAY","applied_by":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]}},{"label":"Code-Review","status":"MAY","applied_by":{"_account_id":25023,"name":"Jonathan Rosser","email":"jonathan.rosser@rd.bbc.co.uk","username":"jrosser"}},{"label":"Workflow","status":"MAY","applied_by":{"_account_id":25023,"name":"Jonathan Rosser","email":"jonathan.rosser@rd.bbc.co.uk","username":"jrosser"}},{"label":"Backport-Candidate","status":"MAY"}]}],"submit_requirements":[{"name":"Verified","description":"Verified in gate by CI","status":"SATISFIED","is_legacy":false,"submittability_expression_result":{"expression":"label:Verified\u003dMAX AND -label:Verified\u003dMIN","fulfilled":true,"status":"PASS","passing_atoms":["label:Verified\u003dMAX"],"failing_atoms":["label:Verified\u003dMIN"],"atom_explanations":{}}},{"name":"Backport-Candidate","description":"Backport candidate status","status":"NOT_APPLICABLE","is_legacy":false,"applicability_expression_result":{"fulfilled":false,"status":"FAIL"},"submittability_expression_result":{"expression":"is:true","fulfilled":true,"status":"NOT_EVALUATED","passing_atoms":[],"failing_atoms":[],"atom_explanations":{}}},{"name":"Code-Review","description":"Code reviewed by core reviewer","status":"SATISFIED","is_legacy":false,"submittability_expression_result":{"expression":"label:Code-Review\u003dMAX AND -label:Code-Review\u003dMIN","fulfilled":true,"status":"PASS","passing_atoms":["label:Code-Review\u003dMAX"],"failing_atoms":["label:Code-Review\u003dMIN"],"atom_explanations":{}}},{"name":"Workflow","description":"Approved for gate by core reviewer","status":"SATISFIED","is_legacy":false,"submittability_expression_result":{"expression":"label:Workflow\u003dMAX AND -label:Workflow\u003dMIN","fulfilled":true,"status":"PASS","passing_atoms":["label:Workflow\u003dMAX"],"failing_atoms":["label:Workflow\u003dMIN"],"atom_explanations":{}}}]}
