)]}'
{"id":"openstack%2Fopenstack-ansible-os_adjutant~985571","triplet_id":"openstack%2Fopenstack-ansible-os_adjutant~master~I7fa3d1c80993ace2d9dba3efa2ba12b3174cbd2c","project":"openstack/openstack-ansible-os_adjutant","branch":"master","topic":"openstack/openstack-ansible-sync/sync-files","hashtags":[],"change_id":"I7fa3d1c80993ace2d9dba3efa2ba12b3174cbd2c","subject":"Updated from OpenStack Ansible Integrated repository","status":"MERGED","created":"2026-04-20 23:14:41.000000000","updated":"2026-04-21 10:40:09.000000000","submitted":"2026-04-21 10:37:54.000000000","submitter":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"total_comment_count":0,"unresolved_comment_count":0,"has_review_started":true,"submission_id":"985571-openstack/openstack-ansible-sync/sync-files","meta_rev_id":"e5f780c8b7e57fd4e51b5f9a430105b534d242ca","_number":985571,"virtual_id_number":985571,"owner":{"_account_id":11131,"name":"OpenStack Proposal Bot","username":"proposal-bot","tags":["SERVICE_USER"]},"actions":{},"labels":{"Verified":{"approved":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"all":[{"value":0,"_account_id":28619,"name":"Dmitriy Rabotyagov","email":"noonedeadpunk@gmail.com","username":"noonedeadpunk"},{"tag":"autogenerated:zuul:gate","value":2,"date":"2026-04-21 10:37:54.000000000","permitted_voting_range":{"min":2,"max":2},"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]}],"values":{"-2":"Fails","-1":"Doesn\u0027t seem to work"," 0":"No score","+1":"Works for me","+2":"Verified"},"description":"","default_value":0,"optional":true},"Code-Review":{"approved":{"_account_id":28619,"name":"Dmitriy Rabotyagov","email":"noonedeadpunk@gmail.com","username":"noonedeadpunk"},"all":[{"value":2,"date":"2026-04-21 08:14:59.000000000","permitted_voting_range":{"min":2,"max":2},"_account_id":28619,"name":"Dmitriy Rabotyagov","email":"noonedeadpunk@gmail.com","username":"noonedeadpunk"},{"value":0,"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]}],"values":{"-2":"Do not merge","-1":"This patch needs further work before it can be merged"," 0":"No score","+1":"Looks good to me, but someone else must approve","+2":"Looks good to me (core reviewer)"},"description":"","default_value":0,"optional":true},"Workflow":{"approved":{"_account_id":28619,"name":"Dmitriy Rabotyagov","email":"noonedeadpunk@gmail.com","username":"noonedeadpunk"},"all":[{"value":1,"date":"2026-04-21 08:14:59.000000000","permitted_voting_range":{"min":1,"max":1},"_account_id":28619,"name":"Dmitriy Rabotyagov","email":"noonedeadpunk@gmail.com","username":"noonedeadpunk"},{"value":0,"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]}],"values":{"-1":"Work in progress"," 0":"Ready for reviews","+1":"Approved"},"description":"","default_value":0,"optional":true},"Backport-Candidate":{"all":[{"value":0,"_account_id":28619,"name":"Dmitriy Rabotyagov","email":"noonedeadpunk@gmail.com","username":"noonedeadpunk"},{"value":0,"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]}],"values":{"-1":"Not A Backport Candidate"," 0":"Backport Review Needed","+1":"Proposed Backport"},"description":"","default_value":0,"optional":true}},"removable_reviewers":[],"reviewers":{"REVIEWER":[{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},{"_account_id":28619,"name":"Dmitriy Rabotyagov","email":"noonedeadpunk@gmail.com","username":"noonedeadpunk"}]},"pending_reviewers":{},"reviewer_updates":[{"updated":"2026-04-21 02:46:45.000000000","updated_by":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"reviewer":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"state":"REVIEWER"},{"updated":"2026-04-21 08:14:59.000000000","updated_by":{"_account_id":28619,"name":"Dmitriy Rabotyagov","email":"noonedeadpunk@gmail.com","username":"noonedeadpunk"},"reviewer":{"_account_id":28619,"name":"Dmitriy Rabotyagov","email":"noonedeadpunk@gmail.com","username":"noonedeadpunk"},"state":"REVIEWER"}],"messages":[{"id":"039b6b24043d403be4aa90fc6c4a750275d6bc77","tag":"autogenerated:gerrit:newPatchSet","author":{"_account_id":11131,"name":"OpenStack Proposal Bot","username":"proposal-bot","tags":["SERVICE_USER"]},"date":"2026-04-20 23:14:41.000000000","message":"Uploaded patch set 1.","accounts_in_message":[],"_revision_number":1},{"id":"84d8f3251668a4f05109e58c39db97b93b6d334f","tag":"autogenerated:zuul:check","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2026-04-21 02:46:45.000000000","message":"Patch Set 1: Verified+1\n\nBuild succeeded (check pipeline).\nhttps://zuul.opendev.org/t/openstack/buildset/4677387674974c35b3ab58c0ec03920e\n\n- openstack-ansible-linters-ubuntu-noble https://zuul.opendev.org/t/openstack/build/12ef19018b1642c19da74702009a72cc : SUCCESS in 12m 45s\n- openstack-ansible-deploy-aio_metal-centos-10-stream https://zuul.opendev.org/t/openstack/build/0c7f5e15f22c41468dc2a9d284642b33 : SUCCESS in 1h 21m 31s\n- openstack-ansible-deploy-aio_metal-debian-bookworm https://zuul.opendev.org/t/openstack/build/16ccfad980564484a266a65720ba82b3 : SUCCESS in 1h 30m 58s\n- openstack-ansible-deploy-aio_metal-debian-trixie https://zuul.opendev.org/t/openstack/build/f9d849d0e1b9421a8d8f9bf8680e191f : SUCCESS in 1h 38m 50s\n- openstack-ansible-deploy-aio_metal-rockylinux-10 https://zuul.opendev.org/t/openstack/build/93ee63dc94ac4d639213db3cfda5533a : SUCCESS in 1h 39m 15s\n- openstack-ansible-deploy-aio_metal_tls-rockylinux-10 https://zuul.opendev.org/t/openstack/build/e237d838c1f74f54a056f19d0e247834 : SUCCESS in 1h 30m 57s\n- openstack-ansible-deploy-aio_metal-ubuntu-noble https://zuul.opendev.org/t/openstack/build/a8be5f32bbbd40f6a5a3d84720c7c659 : SUCCESS in 1h 37m 15s\n- openstack-ansible-deploy-aio_metal_tls-ubuntu-noble https://zuul.opendev.org/t/openstack/build/13803e46d64e4854b93464be05e83bd3 : SUCCESS in 1h 48m 53s\n- openstack-ansible-upgrade-aio_metal-rockylinux-10 https://zuul.opendev.org/t/openstack/build/c990ee75ff3a4c119abd3ab2a1d17021 : SUCCESS in 1h 16m 56s\n- openstack-ansible-upgrade_2025.1-aio_metal-rockylinux-10 https://zuul.opendev.org/t/openstack/build/781ae63f306445e2abc99cab9eac95b2 : SUCCESS in 1h 16m 35s\n- openstack-ansible-upgrade-aio_metal-ubuntu-noble https://zuul.opendev.org/t/openstack/build/83f55887f1ee427e985770a3e59aea0a : SUCCESS in 1h 56m 16s\n- openstack-ansible-upgrade_2025.1-aio_metal-ubuntu-noble https://zuul.opendev.org/t/openstack/build/8d104ad4722d4afbb4daa618c688a1e7 : SUCCESS in 2h 30m 13s\n- openstack-ansible-upgrade-aio_metal_tls-ubuntu-noble https://zuul.opendev.org/t/openstack/build/44cce799465e47139e4d6ad0389f0866 : SUCCESS in 1h 54m 11s\n- openstack-tox-docs https://zuul.opendev.org/t/openstack/build/047305e5d8e142438a9a49aea76cc77c : SUCCESS in 6m 19s\n- build-openstack-releasenotes https://zuul.opendev.org/t/openstack/build/1802a6e70c814ea1af2d088ff8235cd9 : SUCCESS in 3m 24s","accounts_in_message":[],"_revision_number":1},{"id":"42949ccefed6950624e4ae925dd12e652720ca77","author":{"_account_id":28619,"name":"Dmitriy Rabotyagov","email":"noonedeadpunk@gmail.com","username":"noonedeadpunk"},"date":"2026-04-21 08:14:59.000000000","message":"Patch Set 1: Code-Review+2 Workflow+1","accounts_in_message":[],"_revision_number":1},{"id":"97a65c050fa351ead6acb6070e9b83c6c04aa031","tag":"autogenerated:zuul:gate","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2026-04-21 08:22:11.000000000","message":"Patch Set 1: -Verified\n\nStarting gate jobs.","accounts_in_message":[],"_revision_number":1},{"id":"08700c39098647abaf0a39af5f010b0a287e57a9","tag":"autogenerated:zuul:gate","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2026-04-21 10:37:54.000000000","message":"Patch Set 1: Verified+2\n\nBuild succeeded (gate pipeline).\nhttps://zuul.opendev.org/t/openstack/buildset/a106970f26fd45e9bf65d32ce2a2fad9\n\n- openstack-ansible-linters-ubuntu-noble https://zuul.opendev.org/t/openstack/build/97c642e61e1e4962beb3d6d5fe025aa6 : SUCCESS in 12m 42s\n- openstack-ansible-deploy-aio_metal-centos-10-stream https://zuul.opendev.org/t/openstack/build/04d647dc5d2c4f68a1c55063cff68712 : SUCCESS in 1h 02m 08s\n- openstack-ansible-deploy-aio_metal-debian-bookworm https://zuul.opendev.org/t/openstack/build/af46ee3583e04231afd74d6a28538cdd : SUCCESS in 1h 29m 32s\n- openstack-ansible-deploy-aio_metal-debian-trixie https://zuul.opendev.org/t/openstack/build/cf9cd5aa04494a8285bea044a098c3bc : SUCCESS in 1h 28m 50s\n- openstack-ansible-deploy-aio_metal-rockylinux-10 https://zuul.opendev.org/t/openstack/build/c8c8479ae1934b90a9df9bc7268bd6fa : SUCCESS in 1h 20m 49s\n- openstack-ansible-deploy-aio_metal_tls-rockylinux-10 https://zuul.opendev.org/t/openstack/build/05bc41f460c94ac7872648aa4f08222d : SUCCESS in 1h 44m 02s\n- openstack-ansible-deploy-aio_metal-ubuntu-noble https://zuul.opendev.org/t/openstack/build/a211705320f34b04979bc4e1aafbd50f : SUCCESS in 1h 34m 59s\n- openstack-ansible-deploy-aio_metal_tls-ubuntu-noble https://zuul.opendev.org/t/openstack/build/a87f7a40a5074cbb92d51fd3f6c3cdf7 : SUCCESS in 1h 43m 28s\n- openstack-tox-docs https://zuul.opendev.org/t/openstack/build/7e16ef4599274c67b1e0e1fcfd4ca204 : SUCCESS in 6m 38s\n- build-openstack-releasenotes https://zuul.opendev.org/t/openstack/build/2fedd442c1ac4a828f3bc2b705cf0be6 : SUCCESS in 2m 54s","accounts_in_message":[],"_revision_number":1},{"id":"7cdad4bea604f3d670f50cca44368696b90d3de7","tag":"autogenerated:gerrit:merged","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2026-04-21 10:37:54.000000000","message":"Change has been successfully merged","accounts_in_message":[],"_revision_number":1},{"id":"e5f780c8b7e57fd4e51b5f9a430105b534d242ca","tag":"autogenerated:zuul:promote","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2026-04-21 10:40:09.000000000","message":"Patch Set 1:\n\nBuild succeeded (promote pipeline).\nhttps://zuul.opendev.org/t/openstack/buildset/bfc1c9cf70ed419cad4a05fd60421bdc\n\n- promote-openstack-tox-docs https://zuul.opendev.org/t/openstack/build/d642bfb6325b4971ac354ddcdd1b06e0 : SUCCESS in 1m 23s","accounts_in_message":[],"_revision_number":1}],"current_revision_number":1,"current_revision":"a36a5f51842043034f12cf7e4b6774f62865db61","revisions":{"a36a5f51842043034f12cf7e4b6774f62865db61":{"kind":"REWORK","_number":1,"created":"2026-04-20 23:14:41.000000000","uploader":{"_account_id":11131,"name":"OpenStack Proposal Bot","username":"proposal-bot","tags":["SERVICE_USER"]},"ref":"refs/changes/71/985571/1","fetch":{"anonymous http":{"url":"https://review.opendev.org/openstack/openstack-ansible-os_adjutant","ref":"refs/changes/71/985571/1","commands":{"Checkout":"git fetch https://review.opendev.org/openstack/openstack-ansible-os_adjutant refs/changes/71/985571/1 \u0026\u0026 git checkout FETCH_HEAD","Cherry Pick":"git fetch https://review.opendev.org/openstack/openstack-ansible-os_adjutant refs/changes/71/985571/1 \u0026\u0026 git cherry-pick FETCH_HEAD","Format Patch":"git fetch https://review.opendev.org/openstack/openstack-ansible-os_adjutant refs/changes/71/985571/1 \u0026\u0026 git format-patch -1 --stdout FETCH_HEAD","Pull":"git pull https://review.opendev.org/openstack/openstack-ansible-os_adjutant refs/changes/71/985571/1"}}},"commit":{"parents":[{"commit":"53515b7f4d82fa111e3b6404ecb8aa86eb2c1fe5","subject":"Merge \"Fix for OpenStack writing style\"","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/openstack-ansible-os_adjutant/commit/53515b7f4d82fa111e3b6404ecb8aa86eb2c1fe5"}]}],"author":{"name":"OpenStack Proposal Bot","email":"openstack-infra@lists.openstack.org","date":"2026-04-20 23:14:40.000000000","tz":0},"committer":{"name":"OpenStack Proposal Bot","email":"openstack-infra@lists.openstack.org","date":"2026-04-20 23:14:40.000000000","tz":0},"subject":"Updated from OpenStack Ansible Integrated repository","message":"Updated from OpenStack Ansible Integrated repository\n\nChange-Id: I7fa3d1c80993ace2d9dba3efa2ba12b3174cbd2c\nSigned-off-by: OpenStack Proposal Bot \u003copenstack-infra@lists.openstack.org\u003e\nGenerated-By: openstack/project-config:playbooks/proposal/propose_update.sh\n","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/openstack-ansible-os_adjutant/commit/a36a5f51842043034f12cf7e4b6774f62865db61"}],"resolve_conflicts_web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/openstack-ansible-os_adjutant/commit/a36a5f51842043034f12cf7e4b6774f62865db61"}]},"branch":"refs/heads/master"}},"requirements":[],"submit_records":[{"rule_name":"gerrit~DefaultSubmitRule","status":"CLOSED","labels":[{"label":"Verified","status":"MAY","applied_by":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]}},{"label":"Code-Review","status":"MAY","applied_by":{"_account_id":28619,"name":"Dmitriy Rabotyagov","email":"noonedeadpunk@gmail.com","username":"noonedeadpunk"}},{"label":"Workflow","status":"MAY","applied_by":{"_account_id":28619,"name":"Dmitriy Rabotyagov","email":"noonedeadpunk@gmail.com","username":"noonedeadpunk"}},{"label":"Backport-Candidate","status":"MAY"}]}],"submit_requirements":[{"name":"Verified","description":"Verified in gate by CI","status":"SATISFIED","is_legacy":false,"submittability_expression_result":{"expression":"label:Verified\u003dMAX AND -label:Verified\u003dMIN","fulfilled":true,"status":"PASS","passing_atoms":["label:Verified\u003dMAX"],"failing_atoms":["label:Verified\u003dMIN"],"atom_explanations":{"label:Verified\u003dMAX":"","label:Verified\u003dMIN":""}}},{"name":"Backport-Candidate","description":"Backport candidate status","status":"NOT_APPLICABLE","is_legacy":false,"applicability_expression_result":{"fulfilled":false,"status":"FAIL"},"submittability_expression_result":{"expression":"is:true","fulfilled":true,"status":"NOT_EVALUATED","passing_atoms":[],"failing_atoms":[],"atom_explanations":{}}},{"name":"Code-Review","description":"Code reviewed by core reviewer","status":"SATISFIED","is_legacy":false,"submittability_expression_result":{"expression":"label:Code-Review\u003dMAX AND -label:Code-Review\u003dMIN","fulfilled":true,"status":"PASS","passing_atoms":["label:Code-Review\u003dMAX"],"failing_atoms":["label:Code-Review\u003dMIN"],"atom_explanations":{"label:Code-Review\u003dMAX":"","label:Code-Review\u003dMIN":""}}},{"name":"Workflow","description":"Approved for gate by core reviewer","status":"SATISFIED","is_legacy":false,"submittability_expression_result":{"expression":"label:Workflow\u003dMAX AND -label:Workflow\u003dMIN","fulfilled":true,"status":"PASS","passing_atoms":["label:Workflow\u003dMAX"],"failing_atoms":["label:Workflow\u003dMIN"],"atom_explanations":{"label:Workflow\u003dMAX":"","label:Workflow\u003dMIN":""}}}]}
