)]}'
{"id":"openstack%2Fopenstack-ansible-os_aodh~606824","triplet_id":"openstack%2Fopenstack-ansible-os_aodh~master~I92dcc6339f6877f3edffe1c2ce919c5a38b47fa5","project":"openstack/openstack-ansible-os_aodh","branch":"master","topic":"bug/1801657","hashtags":[],"change_id":"I92dcc6339f6877f3edffe1c2ce919c5a38b47fa5","subject":"fix tox python3 overrides","status":"MERGED","created":"2018-09-30 08:24:47.000000000","updated":"2019-04-24 19:27:30.000000000","submitted":"2019-04-24 19:27:30.000000000","submitter":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"total_comment_count":0,"unresolved_comment_count":0,"has_review_started":true,"submission_id":"606824-1556134050816-40cc56dc","meta_rev_id":"d9360c3f871513175d9b63df6d3159a1594ee302","_number":606824,"virtual_id_number":606824,"owner":{"_account_id":26297,"name":"pippo","email":"1211099771@qq.com","username":"huang.zhiping"},"actions":{},"labels":{"Verified":{"approved":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"all":[{"value":0,"_account_id":28935,"name":"98k","email":"18552437190@163.com","username":"98k"},{"value":0,"_account_id":26297,"name":"pippo","email":"1211099771@qq.com","username":"huang.zhiping"},{"value":0,"_account_id":17130,"name":"melissaml","email":"ma.lei@99cloud.net","username":"malei"},{"value":0,"_account_id":17499,"name":"Filippo Inzaghi","email":"zhao.daqing@99cloud.net","username":"qingszhao"},{"value":0,"_account_id":17068,"name":"Jean-Philippe Evrard","email":"openstack@a.spamming.party","username":"evrardjp"},{"value":2,"date":"2019-04-24 19:27:30.000000000","permitted_voting_range":{"min":2,"max":2},"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},{"value":0,"_account_id":1004,"name":"Mohammed Naser","email":"mnaser@vexxhost.com","username":"mnaser"},{"value":0,"_account_id":26285,"name":"wu.chunyang","email":"wchy1001@gmail.com","username":"wu.chunyang"},{"value":0,"_account_id":28543,"name":"shupeng","email":"15050873171@163.com","username":"Vieri"}],"values":{"-2":"Fails","-1":"Doesn\u0027t seem to work"," 0":"No score","+1":"Works for me","+2":"Verified"},"description":"","default_value":0,"optional":true},"Code-Review":{"approved":{"_account_id":17068,"name":"Jean-Philippe Evrard","email":"openstack@a.spamming.party","username":"evrardjp"},"recommended":{"_account_id":17499,"name":"Filippo Inzaghi","email":"zhao.daqing@99cloud.net","username":"qingszhao"},"all":[{"value":1,"date":"2019-01-11 06:09:35.000000000","permitted_voting_range":{"min":1,"max":1},"_account_id":28935,"name":"98k","email":"18552437190@163.com","username":"98k"},{"value":0,"_account_id":26297,"name":"pippo","email":"1211099771@qq.com","username":"huang.zhiping"},{"value":1,"date":"2019-01-11 06:09:35.000000000","permitted_voting_range":{"min":1,"max":1},"_account_id":17130,"name":"melissaml","email":"ma.lei@99cloud.net","username":"malei"},{"value":1,"date":"2019-01-14 13:52:02.000000000","permitted_voting_range":{"min":1,"max":1},"_account_id":17499,"name":"Filippo Inzaghi","email":"zhao.daqing@99cloud.net","username":"qingszhao"},{"value":2,"date":"2019-01-11 06:09:35.000000000","_account_id":17068,"name":"Jean-Philippe Evrard","email":"openstack@a.spamming.party","username":"evrardjp"},{"value":0,"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},{"value":2,"date":"2019-04-24 18:51:27.000000000","permitted_voting_range":{"min":2,"max":2},"_account_id":1004,"name":"Mohammed Naser","email":"mnaser@vexxhost.com","username":"mnaser"},{"value":1,"date":"2019-01-11 06:09:35.000000000","permitted_voting_range":{"min":1,"max":1},"_account_id":26285,"name":"wu.chunyang","email":"wchy1001@gmail.com","username":"wu.chunyang"},{"value":1,"date":"2019-01-14 12:02:42.000000000","permitted_voting_range":{"min":1,"max":1},"_account_id":28543,"name":"shupeng","email":"15050873171@163.com","username":"Vieri"}],"values":{"-2":"Do not merge","-1":"This patch needs further work before it can be merged"," 0":"No score","+1":"Looks good to me, but someone else must approve","+2":"Looks good to me (core reviewer)"},"description":"","value":1,"default_value":0,"optional":true},"Workflow":{"approved":{"_account_id":1004,"name":"Mohammed Naser","email":"mnaser@vexxhost.com","username":"mnaser"},"all":[{"value":0,"_account_id":28935,"name":"98k","email":"18552437190@163.com","username":"98k"},{"value":0,"_account_id":26297,"name":"pippo","email":"1211099771@qq.com","username":"huang.zhiping"},{"value":0,"_account_id":17130,"name":"melissaml","email":"ma.lei@99cloud.net","username":"malei"},{"value":0,"_account_id":17499,"name":"Filippo Inzaghi","email":"zhao.daqing@99cloud.net","username":"qingszhao"},{"value":0,"_account_id":17068,"name":"Jean-Philippe Evrard","email":"openstack@a.spamming.party","username":"evrardjp"},{"value":0,"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},{"value":1,"date":"2019-04-24 18:51:27.000000000","permitted_voting_range":{"min":1,"max":1},"_account_id":1004,"name":"Mohammed Naser","email":"mnaser@vexxhost.com","username":"mnaser"},{"value":0,"_account_id":26285,"name":"wu.chunyang","email":"wchy1001@gmail.com","username":"wu.chunyang"},{"value":0,"_account_id":28543,"name":"shupeng","email":"15050873171@163.com","username":"Vieri"}],"values":{"-1":"Work in progress"," 0":"Ready for reviews","+1":"Approved"},"description":"","default_value":0,"optional":true},"Backport-Candidate":{"all":[{"value":0,"_account_id":28935,"name":"98k","email":"18552437190@163.com","username":"98k"},{"value":0,"_account_id":26297,"name":"pippo","email":"1211099771@qq.com","username":"huang.zhiping"},{"value":0,"_account_id":17130,"name":"melissaml","email":"ma.lei@99cloud.net","username":"malei"},{"value":0,"_account_id":17499,"name":"Filippo Inzaghi","email":"zhao.daqing@99cloud.net","username":"qingszhao"},{"value":0,"_account_id":17068,"name":"Jean-Philippe Evrard","email":"openstack@a.spamming.party","username":"evrardjp"},{"value":0,"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},{"value":0,"_account_id":1004,"name":"Mohammed Naser","email":"mnaser@vexxhost.com","username":"mnaser"},{"value":0,"_account_id":26285,"name":"wu.chunyang","email":"wchy1001@gmail.com","username":"wu.chunyang"},{"value":0,"_account_id":28543,"name":"shupeng","email":"15050873171@163.com","username":"Vieri"}],"values":{"-1":"Not A Backport Candidate"," 0":"Backport Review Needed","+1":"Proposed Backport"},"description":"","default_value":0,"optional":true}},"removable_reviewers":[],"reviewers":{"REVIEWER":[{"_account_id":1004,"name":"Mohammed Naser","email":"mnaser@vexxhost.com","username":"mnaser"},{"_account_id":17068,"name":"Jean-Philippe Evrard","email":"openstack@a.spamming.party","username":"evrardjp"},{"_account_id":17130,"name":"melissaml","email":"ma.lei@99cloud.net","username":"malei"},{"_account_id":17499,"name":"Filippo Inzaghi","email":"zhao.daqing@99cloud.net","username":"qingszhao"},{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},{"_account_id":26285,"name":"wu.chunyang","email":"wchy1001@gmail.com","username":"wu.chunyang"},{"_account_id":26297,"name":"pippo","email":"1211099771@qq.com","username":"huang.zhiping"},{"_account_id":28543,"name":"shupeng","email":"15050873171@163.com","username":"Vieri"},{"_account_id":28935,"name":"98k","email":"18552437190@163.com","username":"98k"}]},"pending_reviewers":{},"reviewer_updates":[{"updated":"2018-11-12 07:08:44.000000000","updated_by":{"_account_id":28935,"name":"98k","email":"18552437190@163.com","username":"98k"},"reviewer":{"_account_id":28935,"name":"98k","email":"18552437190@163.com","username":"98k"},"state":"REVIEWER"},{"updated":"2018-11-12 15:24:12.000000000","updated_by":{"_account_id":17068,"name":"Jean-Philippe Evrard","email":"openstack@a.spamming.party","username":"evrardjp"},"reviewer":{"_account_id":17068,"name":"Jean-Philippe Evrard","email":"openstack@a.spamming.party","username":"evrardjp"},"state":"REVIEWER"},{"updated":"2018-11-12 17:24:43.000000000","updated_by":{"_account_id":17130,"name":"melissaml","email":"ma.lei@99cloud.net","username":"malei"},"reviewer":{"_account_id":17130,"name":"melissaml","email":"ma.lei@99cloud.net","username":"malei"},"state":"REVIEWER"},{"updated":"2018-12-13 14:44:12.000000000","updated_by":{"_account_id":26285,"name":"wu.chunyang","email":"wchy1001@gmail.com","username":"wu.chunyang"},"reviewer":{"_account_id":26285,"name":"wu.chunyang","email":"wchy1001@gmail.com","username":"wu.chunyang"},"state":"REVIEWER"},{"updated":"2019-01-14 12:02:42.000000000","updated_by":{"_account_id":28543,"name":"shupeng","email":"15050873171@163.com","username":"Vieri"},"reviewer":{"_account_id":28543,"name":"shupeng","email":"15050873171@163.com","username":"Vieri"},"state":"REVIEWER"},{"updated":"2019-01-14 13:52:02.000000000","updated_by":{"_account_id":17499,"name":"Filippo Inzaghi","email":"zhao.daqing@99cloud.net","username":"qingszhao"},"reviewer":{"_account_id":17499,"name":"Filippo Inzaghi","email":"zhao.daqing@99cloud.net","username":"qingszhao"},"state":"REVIEWER"},{"updated":"2019-04-24 18:51:27.000000000","updated_by":{"_account_id":1004,"name":"Mohammed Naser","email":"mnaser@vexxhost.com","username":"mnaser"},"reviewer":{"_account_id":1004,"name":"Mohammed Naser","email":"mnaser@vexxhost.com","username":"mnaser"},"state":"REVIEWER"},{"updated":"2019-04-24 19:27:30.000000000","updated_by":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"reviewer":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"state":"REVIEWER"}],"messages":[{"id":"750ea8ff48331c4e2572d10cbf79b8807ec54195","author":{"_account_id":26297,"name":"pippo","email":"1211099771@qq.com","username":"huang.zhiping"},"date":"2018-09-30 08:24:47.000000000","message":"Uploaded patch set 1.","accounts_in_message":[],"_revision_number":1},{"id":"2b4a34be324a4c3176370aa5b78eb7a55c877f00","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2018-09-30 09:24:22.000000000","message":"Patch Set 1: Verified-1\n\nBuild failed (check pipeline).  For information on how to proceed, see\nhttp://docs.openstack.org/infra/manual/developers.html#automated-testing\n\n\n- build-openstack-sphinx-docs http://logs.openstack.org/24/606824/1/check/build-openstack-sphinx-docs/fcef4fa/html/ : SUCCESS in 2m 44s\n- build-openstack-releasenotes http://logs.openstack.org/24/606824/1/check/build-openstack-releasenotes/1a97dfa/html/ : SUCCESS in 2m 36s\n- openstack-ansible-linters http://logs.openstack.org/24/606824/1/check/openstack-ansible-linters/3098e99/ : SUCCESS in 6m 09s\n- openstack-ansible-functional-centos-7 http://logs.openstack.org/24/606824/1/check/openstack-ansible-functional-centos-7/543bcc2/ : SUCCESS in 35m 48s\n- openstack-ansible-functional-opensuse-150 http://logs.openstack.org/24/606824/1/check/openstack-ansible-functional-opensuse-150/47698a4/ : SUCCESS in 44m 44s (non-voting)\n- openstack-ansible-functional-opensuse-423 http://logs.openstack.org/24/606824/1/check/openstack-ansible-functional-opensuse-423/37b5af1/ : SUCCESS in 33m 48s\n- openstack-ansible-functional-ubuntu-bionic http://logs.openstack.org/24/606824/1/check/openstack-ansible-functional-ubuntu-bionic/970d324/ : SUCCESS in 32m 31s\n- openstack-ansible-functional-ubuntu-xenial http://logs.openstack.org/24/606824/1/check/openstack-ansible-functional-ubuntu-xenial/e0a2f4f/ : SUCCESS in 31m 07s\n- openstack-tox-docs http://logs.openstack.org/24/606824/1/check/openstack-tox-docs/bd1ab7c/html/ : SUCCESS in 2m 46s\n- openstack-ansible-functional-distro_install-ubuntu-xenial http://logs.openstack.org/24/606824/1/check/openstack-ansible-functional-distro_install-ubuntu-xenial/dba04bf/ : SUCCESS in 32m 39s\n- openstack-ansible-functional-distro_install-centos-7 http://logs.openstack.org/24/606824/1/check/openstack-ansible-functional-distro_install-centos-7/13c3d50/ : SUCCESS in 35m 25s\n- openstack-ansible-functional-distro_install-opensuse-423 http://logs.openstack.org/24/606824/1/check/openstack-ansible-functional-distro_install-opensuse-423/b84c15b/ : FAILURE in 16m 17s","accounts_in_message":[],"_revision_number":1},{"id":"352e3182d15000fc9cb92dfe29530f01f3434aab","author":{"_account_id":28543,"name":"shupeng","email":"15050873171@163.com","username":"Vieri"},"date":"2018-10-07 09:26:13.000000000","message":"Patch Set 1: Code-Review+1","accounts_in_message":[],"_revision_number":1},{"id":"9b29c8a590ae89dbba3015201cc829fa2868b9c5","author":{"_account_id":26297,"name":"pippo","email":"1211099771@qq.com","username":"huang.zhiping"},"date":"2018-11-08 13:44:25.000000000","message":"Patch Set 2: Commit message was updated.","accounts_in_message":[],"_revision_number":2},{"id":"24d4ad3831668bb9c8b12c51ea211210bcb1c654","author":{"_account_id":26297,"name":"pippo","email":"1211099771@qq.com","username":"huang.zhiping"},"date":"2018-11-08 13:44:36.000000000","message":"Topic set to bug/1801657","accounts_in_message":[],"_revision_number":2},{"id":"f2edb384cc76ce2f419d46df32bbb94958aa87e1","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2018-11-08 14:31:11.000000000","message":"Patch Set 2: Verified+1\n\nBuild succeeded (check pipeline).\n\n- openstack-ansible-functional-distro_install-centos-7 http://logs.openstack.org/24/606824/2/check/openstack-ansible-functional-distro_install-centos-7/721ca9e/ : SUCCESS in 34m 39s\n- openstack-ansible-functional-distro_install-opensuse-150 http://logs.openstack.org/24/606824/2/check/openstack-ansible-functional-distro_install-opensuse-150/a5515c6/ : FAILURE in 40m 24s (non-voting)\n- openstack-ansible-functional-distro_install-opensuse-423 http://logs.openstack.org/24/606824/2/check/openstack-ansible-functional-distro_install-opensuse-423/99790f4/ : FAILURE in 30m 29s (non-voting)\n- openstack-ansible-functional-distro_install-ubuntu-bionic http://logs.openstack.org/24/606824/2/check/openstack-ansible-functional-distro_install-ubuntu-bionic/854576e/ : FAILURE in 40m 31s (non-voting)\n- openstack-ansible-linters http://logs.openstack.org/24/606824/2/check/openstack-ansible-linters/723dfd3/ : SUCCESS in 6m 20s\n- openstack-ansible-functional-centos-7 http://logs.openstack.org/24/606824/2/check/openstack-ansible-functional-centos-7/9b6f896/ : SUCCESS in 42m 28s\n- openstack-ansible-functional-opensuse-150 http://logs.openstack.org/24/606824/2/check/openstack-ansible-functional-opensuse-150/93fd291/ : SUCCESS in 42m 50s (non-voting)\n- openstack-ansible-functional-opensuse-423 http://logs.openstack.org/24/606824/2/check/openstack-ansible-functional-opensuse-423/a642660/ : SUCCESS in 32m 36s\n- openstack-ansible-functional-ubuntu-bionic http://logs.openstack.org/24/606824/2/check/openstack-ansible-functional-ubuntu-bionic/1584eec/ : SUCCESS in 40m 21s\n- openstack-tox-docs http://logs.openstack.org/24/606824/2/check/openstack-tox-docs/2c73105/html/ : SUCCESS in 2m 37s\n- build-openstack-releasenotes http://logs.openstack.org/24/606824/2/check/build-openstack-releasenotes/810cd06/html/ : SUCCESS in 3m 00s","accounts_in_message":[],"_revision_number":2},{"id":"aa20e2139fd6b3a38007c6c58565e9e2447d3196","author":{"_account_id":28935,"name":"98k","email":"18552437190@163.com","username":"98k"},"date":"2018-11-12 07:08:44.000000000","message":"Patch Set 2: Code-Review+1","accounts_in_message":[],"_revision_number":2},{"id":"80dd54daf17611b15092d8ced3f0806be1edbf7c","author":{"_account_id":28543,"name":"shupeng","email":"15050873171@163.com","username":"Vieri"},"date":"2018-11-12 08:08:15.000000000","message":"Patch Set 2: Code-Review+1","accounts_in_message":[],"_revision_number":2},{"id":"197be74c2814f84ff567237172d025311e8ab220","author":{"_account_id":17068,"name":"Jean-Philippe Evrard","email":"openstack@a.spamming.party","username":"evrardjp"},"date":"2018-11-12 15:24:12.000000000","message":"Patch Set 2: Code-Review+2","accounts_in_message":[],"_revision_number":2},{"id":"3d4e8b9ce844880ecb7f979b98241d2e144ff164","author":{"_account_id":17130,"name":"melissaml","email":"ma.lei@99cloud.net","username":"malei"},"date":"2018-11-12 17:24:43.000000000","message":"Patch Set 2: Code-Review+1","accounts_in_message":[],"_revision_number":2},{"id":"a7b20e2ebe284c1dadeed00508df79f840e7bf99","author":{"_account_id":28935,"name":"98k","email":"18552437190@163.com","username":"98k"},"date":"2018-11-26 07:26:29.000000000","message":"Patch Set 3: Patch Set 2 was rebased","accounts_in_message":[],"_revision_number":3},{"id":"563b372910eff06b1ba8b541d9043b089183953e","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2018-11-26 08:36:15.000000000","message":"Patch Set 3: Verified-1\n\nBuild failed (check pipeline).  For information on how to proceed, see\nhttp://docs.openstack.org/infra/manual/developers.html#automated-testing\n\n\n- openstack-ansible-functional-distro_install-centos-7 http://logs.openstack.org/24/606824/3/check/openstack-ansible-functional-distro_install-centos-7/e57ec10/ : FAILURE in 34m 24s\n- openstack-ansible-functional-distro_install-opensuse-150 http://logs.openstack.org/24/606824/3/check/openstack-ansible-functional-distro_install-opensuse-150/3a8c588/ : FAILURE in 31m 46s (non-voting)\n- openstack-ansible-functional-distro_install-opensuse-423 http://logs.openstack.org/24/606824/3/check/openstack-ansible-functional-distro_install-opensuse-423/d427a0c/ : FAILURE in 32m 10s (non-voting)\n- openstack-ansible-functional-distro_install-ubuntu-bionic http://logs.openstack.org/24/606824/3/check/openstack-ansible-functional-distro_install-ubuntu-bionic/6006c85/ : FAILURE in 1h 07m 33s (non-voting)\n- openstack-ansible-linters http://logs.openstack.org/24/606824/3/check/openstack-ansible-linters/fa4d6c9/ : SUCCESS in 5m 23s\n- openstack-ansible-functional-centos-7 http://logs.openstack.org/24/606824/3/check/openstack-ansible-functional-centos-7/4ff717b/ : SUCCESS in 37m 17s\n- openstack-ansible-functional-opensuse-150 http://logs.openstack.org/24/606824/3/check/openstack-ansible-functional-opensuse-150/08e21c3/ : SUCCESS in 42m 24s (non-voting)\n- openstack-ansible-functional-opensuse-423 http://logs.openstack.org/24/606824/3/check/openstack-ansible-functional-opensuse-423/9dbcf6b/ : SUCCESS in 32m 26s\n- openstack-ansible-functional-ubuntu-bionic http://logs.openstack.org/24/606824/3/check/openstack-ansible-functional-ubuntu-bionic/e035515/ : SUCCESS in 32m 08s\n- openstack-tox-docs http://logs.openstack.org/24/606824/3/check/openstack-tox-docs/efaf78d/html/ : SUCCESS in 3m 28s\n- build-openstack-releasenotes http://logs.openstack.org/24/606824/3/check/build-openstack-releasenotes/7b5205e/html/ : SUCCESS in 2m 42s","accounts_in_message":[],"_revision_number":3},{"id":"55c4909f086af462441510c8d035bf7769501439","author":{"_account_id":28935,"name":"98k","email":"18552437190@163.com","username":"98k"},"date":"2018-11-26 08:36:44.000000000","message":"Patch Set 3:\n\nrecheck","accounts_in_message":[],"_revision_number":3},{"id":"67acbdaadd6da51d7f9093d9bd191451aed69ef8","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2018-11-26 09:41:40.000000000","message":"Patch Set 3:\n\nBuild failed (check pipeline).  For information on how to proceed, see\nhttp://docs.openstack.org/infra/manual/developers.html#automated-testing\n\n\n- openstack-ansible-functional-distro_install-centos-7 http://logs.openstack.org/24/606824/3/check/openstack-ansible-functional-distro_install-centos-7/5e27c89/ : FAILURE in 31m 52s\n- openstack-ansible-functional-distro_install-opensuse-150 http://logs.openstack.org/24/606824/3/check/openstack-ansible-functional-distro_install-opensuse-150/bcb30a3/ : FAILURE in 36m 44s (non-voting)\n- openstack-ansible-functional-distro_install-opensuse-423 http://logs.openstack.org/24/606824/3/check/openstack-ansible-functional-distro_install-opensuse-423/1c67d08/ : FAILURE in 36m 48s (non-voting)\n- openstack-ansible-functional-distro_install-ubuntu-bionic http://logs.openstack.org/24/606824/3/check/openstack-ansible-functional-distro_install-ubuntu-bionic/a1bda3a/ : FAILURE in 1h 00m 07s (non-voting)\n- openstack-ansible-linters http://logs.openstack.org/24/606824/3/check/openstack-ansible-linters/c5b62f4/ : SUCCESS in 5m 53s\n- openstack-ansible-functional-centos-7 http://logs.openstack.org/24/606824/3/check/openstack-ansible-functional-centos-7/3996d22/ : SUCCESS in 38m 30s\n- openstack-ansible-functional-opensuse-150 http://logs.openstack.org/24/606824/3/check/openstack-ansible-functional-opensuse-150/3a631f1/ : SUCCESS in 45m 08s (non-voting)\n- openstack-ansible-functional-opensuse-423 http://logs.openstack.org/24/606824/3/check/openstack-ansible-functional-opensuse-423/15fcc2d/ : SUCCESS in 34m 53s\n- openstack-ansible-functional-ubuntu-bionic http://logs.openstack.org/24/606824/3/check/openstack-ansible-functional-ubuntu-bionic/19f76d9/ : SUCCESS in 34m 53s\n- openstack-tox-docs http://logs.openstack.org/24/606824/3/check/openstack-tox-docs/34561a1/html/ : SUCCESS in 3m 04s\n- build-openstack-releasenotes http://logs.openstack.org/24/606824/3/check/build-openstack-releasenotes/adefd33/html/ : SUCCESS in 3m 08s","accounts_in_message":[],"_revision_number":3},{"id":"328b1eb9fcbcc4f0cf3e36494ac076d706c09527","author":{"_account_id":28543,"name":"shupeng","email":"15050873171@163.com","username":"Vieri"},"date":"2018-11-26 12:11:10.000000000","message":"Patch Set 3: -Code-Review\n\nrecheck","accounts_in_message":[],"_revision_number":3},{"id":"328e59f2265c6bbacde67e266e8a0e989a245472","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2018-11-26 12:56:56.000000000","message":"Patch Set 3:\n\nBuild failed (check pipeline).  For information on how to proceed, see\nhttp://docs.openstack.org/infra/manual/developers.html#automated-testing\n\n\n- openstack-ansible-functional-distro_install-centos-7 http://logs.openstack.org/24/606824/3/check/openstack-ansible-functional-distro_install-centos-7/816627c/ : FAILURE in 33m 31s\n- openstack-ansible-functional-distro_install-opensuse-150 http://logs.openstack.org/24/606824/3/check/openstack-ansible-functional-distro_install-opensuse-150/e2ca752/ : FAILURE in 40m 04s (non-voting)\n- openstack-ansible-functional-distro_install-opensuse-423 http://logs.openstack.org/24/606824/3/check/openstack-ansible-functional-distro_install-opensuse-423/8c9f719/ : FAILURE in 30m 50s (non-voting)\n- openstack-ansible-functional-distro_install-ubuntu-bionic http://logs.openstack.org/24/606824/3/check/openstack-ansible-functional-distro_install-ubuntu-bionic/c1bd71f/ : FAILURE in 34m 32s (non-voting)\n- openstack-ansible-linters http://logs.openstack.org/24/606824/3/check/openstack-ansible-linters/0008f40/ : SUCCESS in 7m 00s\n- openstack-ansible-functional-centos-7 http://logs.openstack.org/24/606824/3/check/openstack-ansible-functional-centos-7/946b8d3/ : SUCCESS in 38m 53s\n- openstack-ansible-functional-opensuse-150 http://logs.openstack.org/24/606824/3/check/openstack-ansible-functional-opensuse-150/c8125e8/ : SUCCESS in 36m 04s (non-voting)\n- openstack-ansible-functional-opensuse-423 http://logs.openstack.org/24/606824/3/check/openstack-ansible-functional-opensuse-423/fcf12e5/ : SUCCESS in 35m 15s\n- openstack-ansible-functional-ubuntu-bionic http://logs.openstack.org/24/606824/3/check/openstack-ansible-functional-ubuntu-bionic/622b4bc/ : SUCCESS in 30m 59s\n- openstack-tox-docs http://logs.openstack.org/24/606824/3/check/openstack-tox-docs/04ba278/html/ : SUCCESS in 3m 05s\n- build-openstack-releasenotes http://logs.openstack.org/24/606824/3/check/build-openstack-releasenotes/1beeb1a/html/ : SUCCESS in 2m 29s","accounts_in_message":[],"_revision_number":3},{"id":"b8aec2d3e98bc7af88a9e72ef081fe33b0fb8e94","author":{"_account_id":28543,"name":"shupeng","email":"15050873171@163.com","username":"Vieri"},"date":"2018-11-27 01:46:20.000000000","message":"Patch Set 3:\n\nrecheck","accounts_in_message":[],"_revision_number":3},{"id":"c7ba6b9cb469552333a8814d34d3b45bd435055e","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2018-11-27 04:02:28.000000000","message":"Patch Set 3:\n\nBuild failed (check pipeline).  For information on how to proceed, see\nhttp://docs.openstack.org/infra/manual/developers.html#automated-testing\n\n\n- openstack-ansible-functional-distro_install-centos-7 http://logs.openstack.org/24/606824/3/check/openstack-ansible-functional-distro_install-centos-7/fe630ff/ : FAILURE in 34m 41s\n- openstack-ansible-functional-distro_install-opensuse-150 http://logs.openstack.org/24/606824/3/check/openstack-ansible-functional-distro_install-opensuse-150/b5a7a3a/ : FAILURE in 35m 52s (non-voting)\n- openstack-ansible-functional-distro_install-opensuse-423 http://logs.openstack.org/24/606824/3/check/openstack-ansible-functional-distro_install-opensuse-423/afad1f8/ : FAILURE in 30m 37s (non-voting)\n- openstack-ansible-functional-distro_install-ubuntu-bionic http://logs.openstack.org/24/606824/3/check/openstack-ansible-functional-distro_install-ubuntu-bionic/6fa8585/ : FAILURE in 31m 43s (non-voting)\n- openstack-ansible-linters http://logs.openstack.org/24/606824/3/check/openstack-ansible-linters/3639d8e/ : SUCCESS in 5m 26s\n- openstack-ansible-functional-centos-7 http://logs.openstack.org/24/606824/3/check/openstack-ansible-functional-centos-7/d5d7fd8/ : SUCCESS in 37m 15s\n- openstack-ansible-functional-opensuse-150 http://logs.openstack.org/24/606824/3/check/openstack-ansible-functional-opensuse-150/3cc53bc/ : SUCCESS in 47m 21s (non-voting)\n- openstack-ansible-functional-opensuse-423 http://logs.openstack.org/24/606824/3/check/openstack-ansible-functional-opensuse-423/1c463ab/ : SUCCESS in 32m 20s\n- openstack-ansible-functional-ubuntu-bionic http://logs.openstack.org/24/606824/3/check/openstack-ansible-functional-ubuntu-bionic/e6f5113/ : SUCCESS in 31m 37s\n- openstack-tox-docs http://logs.openstack.org/24/606824/3/check/openstack-tox-docs/f78afde/html/ : SUCCESS in 2m 50s\n- build-openstack-releasenotes http://logs.openstack.org/24/606824/3/check/build-openstack-releasenotes/90ac513/html/ : SUCCESS in 2m 38s","accounts_in_message":[],"_revision_number":3},{"id":"45cfeac7e67c97492fe8aa87522d039a98a390ec","author":{"_account_id":28543,"name":"shupeng","email":"15050873171@163.com","username":"Vieri"},"date":"2018-11-28 02:03:20.000000000","message":"Patch Set 3:\n\nrecheck","accounts_in_message":[],"_revision_number":3},{"id":"5053b6f56be6544571eaa310c7b00e679d09770c","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2018-11-28 03:48:31.000000000","message":"Patch Set 3:\n\nBuild failed (check pipeline).  For information on how to proceed, see\nhttp://docs.openstack.org/infra/manual/developers.html#automated-testing\n\n\n- openstack-ansible-functional-distro_install-centos-7 http://logs.openstack.org/24/606824/3/check/openstack-ansible-functional-distro_install-centos-7/84b02e2/ : FAILURE in 33m 11s\n- openstack-ansible-functional-distro_install-opensuse-150 http://logs.openstack.org/24/606824/3/check/openstack-ansible-functional-distro_install-opensuse-150/b83e93d/ : FAILURE in 31m 19s (non-voting)\n- openstack-ansible-functional-distro_install-opensuse-423 http://logs.openstack.org/24/606824/3/check/openstack-ansible-functional-distro_install-opensuse-423/2e33c84/ : FAILURE in 28m 28s (non-voting)\n- openstack-ansible-functional-distro_install-ubuntu-bionic http://logs.openstack.org/24/606824/3/check/openstack-ansible-functional-distro_install-ubuntu-bionic/bd2a76f/ : FAILURE in 34m 15s (non-voting)\n- openstack-ansible-linters http://logs.openstack.org/24/606824/3/check/openstack-ansible-linters/0aa5fcb/ : SUCCESS in 8m 43s\n- openstack-ansible-functional-centos-7 http://logs.openstack.org/24/606824/3/check/openstack-ansible-functional-centos-7/9423206/ : SUCCESS in 43m 24s\n- openstack-ansible-functional-opensuse-150 http://logs.openstack.org/24/606824/3/check/openstack-ansible-functional-opensuse-150/5635620/ : TIMED_OUT in 1h 37m 25s (non-voting)\n- openstack-ansible-functional-opensuse-423 http://logs.openstack.org/24/606824/3/check/openstack-ansible-functional-opensuse-423/1badc49/ : SUCCESS in 33m 20s\n- openstack-ansible-functional-ubuntu-bionic http://logs.openstack.org/24/606824/3/check/openstack-ansible-functional-ubuntu-bionic/1963705/ : SUCCESS in 35m 02s\n- openstack-tox-docs http://logs.openstack.org/24/606824/3/check/openstack-tox-docs/c2d2ad0/html/ : SUCCESS in 3m 04s\n- build-openstack-releasenotes http://logs.openstack.org/24/606824/3/check/build-openstack-releasenotes/747b929/html/ : SUCCESS in 2m 32s","accounts_in_message":[],"_revision_number":3},{"id":"0fe2a1277cacc32cd33341dc580425aac3a24046","author":{"_account_id":26297,"name":"pippo","email":"1211099771@qq.com","username":"huang.zhiping"},"date":"2018-11-28 09:13:55.000000000","message":"Patch Set 3:\n\nrecheck","accounts_in_message":[],"_revision_number":3},{"id":"265043775cbfb7aeab43195e29e1f88e465a4bf5","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2018-11-28 11:17:56.000000000","message":"Patch Set 3:\n\nBuild failed (check pipeline).  For information on how to proceed, see\nhttp://docs.openstack.org/infra/manual/developers.html#automated-testing\n\n\n- openstack-ansible-functional-distro_install-centos-7 http://logs.openstack.org/24/606824/3/check/openstack-ansible-functional-distro_install-centos-7/5f8cfc0/ : FAILURE in 22m 25s\n- openstack-ansible-functional-distro_install-opensuse-150 http://logs.openstack.org/24/606824/3/check/openstack-ansible-functional-distro_install-opensuse-150/ebd826b/ : FAILURE in 39m 29s (non-voting)\n- openstack-ansible-functional-distro_install-opensuse-423 http://logs.openstack.org/24/606824/3/check/openstack-ansible-functional-distro_install-opensuse-423/5f3c6da/ : FAILURE in 33m 05s (non-voting)\n- openstack-ansible-functional-distro_install-ubuntu-bionic http://logs.openstack.org/24/606824/3/check/openstack-ansible-functional-distro_install-ubuntu-bionic/51979d4/ : FAILURE in 34m 49s (non-voting)\n- openstack-ansible-linters http://logs.openstack.org/24/606824/3/check/openstack-ansible-linters/36ef7ce/ : TIMED_OUT in 15m 55s\n- openstack-ansible-functional-centos-7 http://logs.openstack.org/24/606824/3/check/openstack-ansible-functional-centos-7/3d39c8e/ : SUCCESS in 39m 48s\n- openstack-ansible-functional-opensuse-150 http://logs.openstack.org/24/606824/3/check/openstack-ansible-functional-opensuse-150/c52ce9f/ : SUCCESS in 1h 03m 32s (non-voting)\n- openstack-ansible-functional-opensuse-423 http://logs.openstack.org/24/606824/3/check/openstack-ansible-functional-opensuse-423/6de4650/ : SUCCESS in 34m 46s\n- openstack-ansible-functional-ubuntu-bionic http://logs.openstack.org/24/606824/3/check/openstack-ansible-functional-ubuntu-bionic/d0f4c36/ : SUCCESS in 34m 00s\n- openstack-tox-docs http://logs.openstack.org/24/606824/3/check/openstack-tox-docs/91b75ac/html/ : SUCCESS in 3m 24s\n- build-openstack-releasenotes http://logs.openstack.org/24/606824/3/check/build-openstack-releasenotes/d59c16b/html/ : SUCCESS in 11m 56s","accounts_in_message":[],"_revision_number":3},{"id":"56c4150500fa9d514926368c61d7a169f86c13a0","author":{"_account_id":28543,"name":"shupeng","email":"15050873171@163.com","username":"Vieri"},"date":"2018-12-07 18:54:43.000000000","message":"Patch Set 3:\n\nrecheck","accounts_in_message":[],"_revision_number":3},{"id":"3cafbfbaa175bc161c41c04b197ea04d4af210a8","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2018-12-07 19:59:47.000000000","message":"Patch Set 3:\n\nBuild failed (check pipeline).  For information on how to proceed, see\nhttp://docs.openstack.org/infra/manual/developers.html#automated-testing\n\n\n- openstack-ansible-functional-distro_install-centos-7 http://logs.openstack.org/24/606824/3/check/openstack-ansible-functional-distro_install-centos-7/5db45dc/ : FAILURE in 32m 26s\n- openstack-ansible-functional-distro_install-opensuse-150 http://logs.openstack.org/24/606824/3/check/openstack-ansible-functional-distro_install-opensuse-150/859e192/ : FAILURE in 36m 21s (non-voting)\n- openstack-ansible-functional-distro_install-opensuse-423 http://logs.openstack.org/24/606824/3/check/openstack-ansible-functional-distro_install-opensuse-423/7c08fa2/ : FAILURE in 28m 56s (non-voting)\n- openstack-ansible-functional-distro_install-ubuntu-bionic http://logs.openstack.org/24/606824/3/check/openstack-ansible-functional-distro_install-ubuntu-bionic/da53e3b/ : SUCCESS in 37m 05s (non-voting)\n- openstack-ansible-linters http://logs.openstack.org/24/606824/3/check/openstack-ansible-linters/7c3089d/ : SUCCESS in 5m 56s\n- openstack-ansible-functional-centos-7 http://logs.openstack.org/24/606824/3/check/openstack-ansible-functional-centos-7/4153af8/ : SUCCESS in 43m 13s\n- openstack-ansible-functional-opensuse-150 http://logs.openstack.org/24/606824/3/check/openstack-ansible-functional-opensuse-150/812d138/ : SUCCESS in 44m 02s (non-voting)\n- openstack-ansible-functional-opensuse-423 http://logs.openstack.org/24/606824/3/check/openstack-ansible-functional-opensuse-423/334070b/ : SUCCESS in 34m 45s\n- openstack-ansible-functional-ubuntu-bionic http://logs.openstack.org/24/606824/3/check/openstack-ansible-functional-ubuntu-bionic/0033c14/ : SUCCESS in 32m 16s\n- openstack-tox-docs http://logs.openstack.org/24/606824/3/check/openstack-tox-docs/d22a249/html/ : SUCCESS in 2m 28s\n- build-openstack-releasenotes http://logs.openstack.org/24/606824/3/check/build-openstack-releasenotes/18b7dd8/html/ : SUCCESS in 3m 09s","accounts_in_message":[],"_revision_number":3},{"id":"1e53e08ed4d6112f094116629f51a00c1dc8e530","author":{"_account_id":26285,"name":"wu.chunyang","email":"wchy1001@gmail.com","username":"wu.chunyang"},"date":"2018-12-13 14:44:12.000000000","message":"Patch Set 3: Code-Review+1","accounts_in_message":[],"_revision_number":3},{"id":"e3098c867bedc9f0da453c46f172de66b46ed761","author":{"_account_id":28543,"name":"shupeng","email":"15050873171@163.com","username":"Vieri"},"date":"2019-01-11 06:09:35.000000000","message":"Patch Set 4: Patch Set 3 was rebased","accounts_in_message":[],"_revision_number":4},{"id":"ab6a4e89d9d5d3f8d9edafe965e88afff3319041","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2019-01-11 07:06:29.000000000","message":"Patch Set 4: Verified-1\n\nBuild failed (check pipeline).  For information on how to proceed, see\nhttp://docs.openstack.org/infra/manual/developers.html#automated-testing\n\n\n- openstack-ansible-functional-distro_install-centos-7 http://logs.openstack.org/24/606824/4/check/openstack-ansible-functional-distro_install-centos-7/6232cf1/ : FAILURE in 34m 07s\n- openstack-ansible-functional-distro_install-opensuse-150 http://logs.openstack.org/24/606824/4/check/openstack-ansible-functional-distro_install-opensuse-150/d007455/ : FAILURE in 23m 38s (non-voting)\n- openstack-ansible-functional-distro_install-opensuse-423 http://logs.openstack.org/24/606824/4/check/openstack-ansible-functional-distro_install-opensuse-423/64190fc/ : FAILURE in 19m 23s (non-voting)\n- openstack-ansible-functional-distro_install-ubuntu-bionic http://logs.openstack.org/24/606824/4/check/openstack-ansible-functional-distro_install-ubuntu-bionic/365d38a/ : SUCCESS in 39m 09s (non-voting)\n- openstack-ansible-linters http://logs.openstack.org/24/606824/4/check/openstack-ansible-linters/cee01f6/ : SUCCESS in 6m 05s\n- openstack-ansible-functional-centos-7 http://logs.openstack.org/24/606824/4/check/openstack-ansible-functional-centos-7/4944c21/ : SUCCESS in 40m 53s\n- openstack-ansible-functional-opensuse-150 http://logs.openstack.org/24/606824/4/check/openstack-ansible-functional-opensuse-150/7323ce4/ : SUCCESS in 45m 21s (non-voting)\n- openstack-ansible-functional-opensuse-423 http://logs.openstack.org/24/606824/4/check/openstack-ansible-functional-opensuse-423/4b8c24f/ : SUCCESS in 38m 21s\n- openstack-ansible-functional-ubuntu-bionic http://logs.openstack.org/24/606824/4/check/openstack-ansible-functional-ubuntu-bionic/718b80b/ : SUCCESS in 34m 44s\n- openstack-tox-docs http://logs.openstack.org/24/606824/4/check/openstack-tox-docs/621c6a7/html/ : SUCCESS in 4m 00s\n- build-openstack-releasenotes http://logs.openstack.org/24/606824/4/check/build-openstack-releasenotes/ae8580b/html/ : SUCCESS in 3m 30s","accounts_in_message":[],"_revision_number":4},{"id":"1fcb258cb32bf89aeb8eff66e2568f06ed6506cd","author":{"_account_id":28543,"name":"shupeng","email":"15050873171@163.com","username":"Vieri"},"date":"2019-01-14 12:02:42.000000000","message":"Patch Set 4: Code-Review+1","accounts_in_message":[],"_revision_number":4},{"id":"33b1096b0691bf4e5d9658416b8268436e63546a","author":{"_account_id":17499,"name":"Filippo Inzaghi","email":"zhao.daqing@99cloud.net","username":"qingszhao"},"date":"2019-01-14 13:52:02.000000000","message":"Patch Set 4: Code-Review+1","accounts_in_message":[],"_revision_number":4},{"id":"e9ad9abc04ee707fce6ebba8b8c0d4004268be5f","author":{"_account_id":1004,"name":"Mohammed Naser","email":"mnaser@vexxhost.com","username":"mnaser"},"date":"2019-04-24 18:51:27.000000000","message":"Patch Set 4: Code-Review+2 Workflow+1\n\nrecheck","accounts_in_message":[],"_revision_number":4},{"id":"eeabd0cca04f5f034d0eb3e7afb2b805da60d995","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2019-04-24 19:04:43.000000000","message":"Patch Set 4: Verified+1\n\nBuild succeeded (check pipeline).\n\n- openstack-ansible-linters http://logs.openstack.org/24/606824/4/check/openstack-ansible-linters/7af11bb/ : SUCCESS in 4m 09s\n- openstack-tox-docs http://logs.openstack.org/24/606824/4/check/openstack-tox-docs/f7dd658/html/ : SUCCESS in 2m 17s\n- build-openstack-releasenotes http://logs.openstack.org/24/606824/4/check/build-openstack-releasenotes/5e729e3/html/ : SUCCESS in 2m 13s","accounts_in_message":[],"_revision_number":4},{"id":"4d8d5d1b702a61ddf0388805b302a3316de53e90","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2019-04-24 19:09:47.000000000","message":"Patch Set 4: -Verified\n\nStarting gate jobs.","accounts_in_message":[],"_revision_number":4},{"id":"b163aaefbd7e6509eab36159e6c7a5c517e25a1e","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2019-04-24 19:27:30.000000000","message":"Patch Set 4: Verified+2\n\nBuild succeeded (gate pipeline).\n\n- openstack-ansible-linters http://logs.openstack.org/24/606824/4/gate/openstack-ansible-linters/c8dd367/ : SUCCESS in 10m 23s\n- openstack-tox-docs http://logs.openstack.org/24/606824/4/gate/openstack-tox-docs/4775f36/html/ : SUCCESS in 2m 18s\n- build-openstack-releasenotes http://logs.openstack.org/24/606824/4/gate/build-openstack-releasenotes/e215ae9/html/ : SUCCESS in 2m 39s","accounts_in_message":[],"_revision_number":4},{"id":"d9360c3f871513175d9b63df6d3159a1594ee302","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2019-04-24 19:27:30.000000000","message":"Change has been successfully merged by Zuul","accounts_in_message":[],"_revision_number":4}],"current_revision_number":4,"current_revision":"ea262b41bcb22af27809642fc1083ded2704cfb5","revisions":{"ee6ee6ff14412bcc760b7e085cac7fac830b10ad":{"kind":"REWORK","_number":1,"created":"2018-09-30 08:24:47.000000000","uploader":{"_account_id":26297,"name":"pippo","email":"1211099771@qq.com","username":"huang.zhiping"},"ref":"refs/changes/24/606824/1","fetch":{"anonymous http":{"url":"https://review.opendev.org/openstack/openstack-ansible-os_aodh","ref":"refs/changes/24/606824/1","commands":{"Checkout":"git fetch https://review.opendev.org/openstack/openstack-ansible-os_aodh refs/changes/24/606824/1 \u0026\u0026 git checkout FETCH_HEAD","Cherry Pick":"git fetch https://review.opendev.org/openstack/openstack-ansible-os_aodh refs/changes/24/606824/1 \u0026\u0026 git cherry-pick FETCH_HEAD","Format Patch":"git fetch https://review.opendev.org/openstack/openstack-ansible-os_aodh refs/changes/24/606824/1 \u0026\u0026 git format-patch -1 --stdout FETCH_HEAD","Pull":"git pull https://review.opendev.org/openstack/openstack-ansible-os_aodh refs/changes/24/606824/1"}}},"commit":{"parents":[{"commit":"b13529ff5dd4bb36e60837489e01a3dd1b783450","subject":"Merge \"fix tox python3 overrides\"","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/openstack-ansible-os_aodh/commit/b13529ff5dd4bb36e60837489e01a3dd1b783450"}]}],"author":{"name":"huang.zhiping","email":"huang.zhiping@99cloud.net","date":"2018-09-30 08:24:32.000000000","tz":0},"committer":{"name":"huang.zhiping","email":"huang.zhiping@99cloud.net","date":"2018-09-30 08:24:47.000000000","tz":0},"subject":"fix tox python3 overrides","message":"fix tox python3 overrides\n\nWe want to default to running all tox environments under python 3, so\nset the basepython value in each environment.\n\nWe do not want to specify a minor version number, because we do not\nwant to have to update the file every time we upgrade python.\n\nWe do not want to set the override once in testenv, because that\nbreaks the more specific versions used in default environments like\npy35 and py36.\n\nChange-Id: I92dcc6339f6877f3edffe1c2ce919c5a38b47fa5\n","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/openstack-ansible-os_aodh/commit/ee6ee6ff14412bcc760b7e085cac7fac830b10ad"}],"resolve_conflicts_web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/openstack-ansible-os_aodh/commit/ee6ee6ff14412bcc760b7e085cac7fac830b10ad"}]},"branch":"refs/heads/master"},"ee674e2e3826901d1e113145aa915435e38c94a9":{"kind":"NO_CODE_CHANGE","_number":2,"created":"2018-11-08 13:44:25.000000000","uploader":{"_account_id":26297,"name":"pippo","email":"1211099771@qq.com","username":"huang.zhiping"},"ref":"refs/changes/24/606824/2","fetch":{"anonymous http":{"url":"https://review.opendev.org/openstack/openstack-ansible-os_aodh","ref":"refs/changes/24/606824/2","commands":{"Checkout":"git fetch https://review.opendev.org/openstack/openstack-ansible-os_aodh refs/changes/24/606824/2 \u0026\u0026 git checkout FETCH_HEAD","Cherry Pick":"git fetch https://review.opendev.org/openstack/openstack-ansible-os_aodh refs/changes/24/606824/2 \u0026\u0026 git cherry-pick FETCH_HEAD","Format Patch":"git fetch https://review.opendev.org/openstack/openstack-ansible-os_aodh refs/changes/24/606824/2 \u0026\u0026 git format-patch -1 --stdout FETCH_HEAD","Pull":"git pull https://review.opendev.org/openstack/openstack-ansible-os_aodh refs/changes/24/606824/2"}}},"commit":{"parents":[{"commit":"b13529ff5dd4bb36e60837489e01a3dd1b783450","subject":"Merge \"fix tox python3 overrides\"","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/openstack-ansible-os_aodh/commit/b13529ff5dd4bb36e60837489e01a3dd1b783450"}]}],"author":{"name":"huang.zhiping","email":"huang.zhiping@99cloud.net","date":"2018-09-30 08:24:32.000000000","tz":0},"committer":{"name":"Andriy Shevchenko","email":"huang.zhiping@99cloud.net","date":"2018-11-08 13:44:20.000000000","tz":0},"subject":"fix tox python3 overrides","message":"fix tox python3 overrides\n\nWe want to default to running all tox environments under python 3, so\nset the basepython value in each environment.\n\nWe do not want to specify a minor version number, because we do not\nwant to have to update the file every time we upgrade python.\n\nWe do not want to set the override once in testenv, because that\nbreaks the more specific versions used in default environments like\npy35 and py36.\n\nChange-Id: I92dcc6339f6877f3edffe1c2ce919c5a38b47fa5\nCloses-Bug:  #1801657\n","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/openstack-ansible-os_aodh/commit/ee674e2e3826901d1e113145aa915435e38c94a9"}],"resolve_conflicts_web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/openstack-ansible-os_aodh/commit/ee674e2e3826901d1e113145aa915435e38c94a9"}]},"branch":"refs/heads/master"},"80227fcb9eb055354bb89dd01ffe4ff9d9526504":{"kind":"TRIVIAL_REBASE","_number":3,"created":"2018-11-26 07:26:29.000000000","uploader":{"_account_id":28935,"name":"98k","email":"18552437190@163.com","username":"98k"},"ref":"refs/changes/24/606824/3","fetch":{"anonymous http":{"url":"https://review.opendev.org/openstack/openstack-ansible-os_aodh","ref":"refs/changes/24/606824/3","commands":{"Checkout":"git fetch https://review.opendev.org/openstack/openstack-ansible-os_aodh refs/changes/24/606824/3 \u0026\u0026 git checkout FETCH_HEAD","Cherry Pick":"git fetch https://review.opendev.org/openstack/openstack-ansible-os_aodh refs/changes/24/606824/3 \u0026\u0026 git cherry-pick FETCH_HEAD","Format Patch":"git fetch https://review.opendev.org/openstack/openstack-ansible-os_aodh refs/changes/24/606824/3 \u0026\u0026 git format-patch -1 --stdout FETCH_HEAD","Pull":"git pull https://review.opendev.org/openstack/openstack-ansible-os_aodh refs/changes/24/606824/3"}}},"commit":{"parents":[{"commit":"d2dfb8cdbd56e095bac09bac3f8eea87361287d9","subject":"Remove unnecessary package install duplication","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/openstack-ansible-os_aodh/commit/d2dfb8cdbd56e095bac09bac3f8eea87361287d9"}]}],"author":{"name":"huang.zhiping","email":"huang.zhiping@99cloud.net","date":"2018-09-30 08:24:32.000000000","tz":0},"committer":{"name":"98k","email":"18552437190@163.com","date":"2018-11-26 07:26:29.000000000","tz":0},"subject":"fix tox python3 overrides","message":"fix tox python3 overrides\n\nWe want to default to running all tox environments under python 3, so\nset the basepython value in each environment.\n\nWe do not want to specify a minor version number, because we do not\nwant to have to update the file every time we upgrade python.\n\nWe do not want to set the override once in testenv, because that\nbreaks the more specific versions used in default environments like\npy35 and py36.\n\nChange-Id: I92dcc6339f6877f3edffe1c2ce919c5a38b47fa5\nCloses-Bug:  #1801657\n","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/openstack-ansible-os_aodh/commit/80227fcb9eb055354bb89dd01ffe4ff9d9526504"}],"resolve_conflicts_web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/openstack-ansible-os_aodh/commit/80227fcb9eb055354bb89dd01ffe4ff9d9526504"}]},"branch":"refs/heads/master"},"ea262b41bcb22af27809642fc1083ded2704cfb5":{"kind":"TRIVIAL_REBASE","_number":4,"created":"2019-01-11 06:09:35.000000000","uploader":{"_account_id":28543,"name":"shupeng","email":"15050873171@163.com","username":"Vieri"},"ref":"refs/changes/24/606824/4","fetch":{"anonymous http":{"url":"https://review.opendev.org/openstack/openstack-ansible-os_aodh","ref":"refs/changes/24/606824/4","commands":{"Checkout":"git fetch https://review.opendev.org/openstack/openstack-ansible-os_aodh refs/changes/24/606824/4 \u0026\u0026 git checkout FETCH_HEAD","Cherry Pick":"git fetch https://review.opendev.org/openstack/openstack-ansible-os_aodh refs/changes/24/606824/4 \u0026\u0026 git cherry-pick FETCH_HEAD","Format Patch":"git fetch https://review.opendev.org/openstack/openstack-ansible-os_aodh refs/changes/24/606824/4 \u0026\u0026 git format-patch -1 --stdout FETCH_HEAD","Pull":"git pull https://review.opendev.org/openstack/openstack-ansible-os_aodh refs/changes/24/606824/4"}}},"commit":{"parents":[{"commit":"966141ddcb4d6140d884f62ebedbb0cdbdda6d3b","subject":"Update mailinglist from dev to discuss","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/openstack-ansible-os_aodh/commit/966141ddcb4d6140d884f62ebedbb0cdbdda6d3b"}]}],"author":{"name":"huang.zhiping","email":"huang.zhiping@99cloud.net","date":"2018-09-30 08:24:32.000000000","tz":0},"committer":{"name":"Vieri","email":"15050873171@163.com","date":"2019-01-11 06:09:35.000000000","tz":0},"subject":"fix tox python3 overrides","message":"fix tox python3 overrides\n\nWe want to default to running all tox environments under python 3, so\nset the basepython value in each environment.\n\nWe do not want to specify a minor version number, because we do not\nwant to have to update the file every time we upgrade python.\n\nWe do not want to set the override once in testenv, because that\nbreaks the more specific versions used in default environments like\npy35 and py36.\n\nChange-Id: I92dcc6339f6877f3edffe1c2ce919c5a38b47fa5\nCloses-Bug:  #1801657\n","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/openstack-ansible-os_aodh/commit/ea262b41bcb22af27809642fc1083ded2704cfb5"}],"resolve_conflicts_web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/openstack-ansible-os_aodh/commit/ea262b41bcb22af27809642fc1083ded2704cfb5"}]},"branch":"refs/heads/master"}},"requirements":[],"submit_records":[],"submit_requirements":[]}
