)]}'
{"id":"openstack%2Fopenstack-ansible-os_aodh~654897","triplet_id":"openstack%2Fopenstack-ansible-os_aodh~stable%2Fstein~I7429f11f46a893ab93982aad81bc2f422f6a8b66","project":"openstack/openstack-ansible-os_aodh","branch":"stable/stein","topic":"openstack/openstack-ansible-tests/sync-tests","hashtags":[],"change_id":"I7429f11f46a893ab93982aad81bc2f422f6a8b66","subject":"Updated from OpenStack Ansible Tests","status":"MERGED","created":"2019-04-23 09:23:37.000000000","updated":"2019-04-24 00:12:00.000000000","submitted":"2019-04-24 00:12:00.000000000","submitter":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"total_comment_count":0,"unresolved_comment_count":0,"has_review_started":true,"submission_id":"654897-1556064720306-332a2ac9","meta_rev_id":"b221d95cf4572b76450de2af4944a9c1b38790b3","_number":654897,"virtual_id_number":654897,"owner":{"_account_id":11131,"name":"OpenStack Proposal Bot","username":"proposal-bot","tags":["SERVICE_USER"]},"actions":{},"labels":{"Verified":{"approved":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"all":[{"value":0,"_account_id":28619,"name":"Dmitriy Rabotyagov","email":"noonedeadpunk@gmail.com","username":"noonedeadpunk"},{"value":2,"date":"2019-04-24 00:11:59.000000000","permitted_voting_range":{"min":2,"max":2},"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},{"value":0,"date":"2019-04-23 20:21:21.000000000","_account_id":25023,"name":"Jonathan Rosser","email":"jonathan.rosser@rd.bbc.co.uk","username":"jrosser"}],"values":{"-2":"Fails","-1":"Doesn\u0027t seem to work"," 0":"No score","+1":"Works for me","+2":"Verified"},"description":"","default_value":0,"optional":true},"Code-Review":{"approved":{"_account_id":28619,"name":"Dmitriy Rabotyagov","email":"noonedeadpunk@gmail.com","username":"noonedeadpunk"},"all":[{"value":2,"date":"2019-04-23 10:52:05.000000000","permitted_voting_range":{"min":2,"max":2},"_account_id":28619,"name":"Dmitriy Rabotyagov","email":"noonedeadpunk@gmail.com","username":"noonedeadpunk"},{"value":0,"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},{"value":0,"_account_id":25023,"name":"Jonathan Rosser","email":"jonathan.rosser@rd.bbc.co.uk","username":"jrosser"}],"values":{"-2":"Do not merge","-1":"This patch needs further work before it can be merged"," 0":"No score","+1":"Looks good to me, but someone else must approve","+2":"Looks good to me (core reviewer)"},"description":"","default_value":0,"optional":true},"Workflow":{"approved":{"_account_id":28619,"name":"Dmitriy Rabotyagov","email":"noonedeadpunk@gmail.com","username":"noonedeadpunk"},"all":[{"value":1,"date":"2019-04-23 10:52:05.000000000","permitted_voting_range":{"min":1,"max":1},"_account_id":28619,"name":"Dmitriy Rabotyagov","email":"noonedeadpunk@gmail.com","username":"noonedeadpunk"},{"value":0,"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},{"value":0,"_account_id":25023,"name":"Jonathan Rosser","email":"jonathan.rosser@rd.bbc.co.uk","username":"jrosser"}],"values":{"-1":"Work in progress"," 0":"Ready for reviews","+1":"Approved"},"description":"","default_value":0,"optional":true},"Backport-Candidate":{"all":[{"value":0,"_account_id":28619,"name":"Dmitriy Rabotyagov","email":"noonedeadpunk@gmail.com","username":"noonedeadpunk"},{"value":0,"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},{"value":0,"_account_id":25023,"name":"Jonathan Rosser","email":"jonathan.rosser@rd.bbc.co.uk","username":"jrosser"}],"values":{"-1":"Not A Backport Candidate"," 0":"Backport Review Needed","+1":"Proposed Backport"},"description":"","default_value":0,"optional":true}},"removable_reviewers":[],"reviewers":{"REVIEWER":[{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},{"_account_id":25023,"name":"Jonathan Rosser","email":"jonathan.rosser@rd.bbc.co.uk","username":"jrosser"},{"_account_id":28619,"name":"Dmitriy Rabotyagov","email":"noonedeadpunk@gmail.com","username":"noonedeadpunk"}]},"pending_reviewers":{},"reviewer_updates":[{"updated":"2019-04-23 10:52:05.000000000","updated_by":{"_account_id":28619,"name":"Dmitriy Rabotyagov","email":"noonedeadpunk@gmail.com","username":"noonedeadpunk"},"reviewer":{"_account_id":28619,"name":"Dmitriy Rabotyagov","email":"noonedeadpunk@gmail.com","username":"noonedeadpunk"},"state":"REVIEWER"},{"updated":"2019-04-23 20:21:21.000000000","updated_by":{"_account_id":25023,"name":"Jonathan Rosser","email":"jonathan.rosser@rd.bbc.co.uk","username":"jrosser"},"reviewer":{"_account_id":25023,"name":"Jonathan Rosser","email":"jonathan.rosser@rd.bbc.co.uk","username":"jrosser"},"state":"REVIEWER"},{"updated":"2019-04-24 00:11:59.000000000","updated_by":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"reviewer":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"state":"REVIEWER"}],"messages":[{"id":"d07928c7463362aea09384927567d10e211d23a9","author":{"_account_id":11131,"name":"OpenStack Proposal Bot","username":"proposal-bot","tags":["SERVICE_USER"]},"date":"2019-04-23 09:23:37.000000000","message":"Uploaded patch set 1.","accounts_in_message":[],"_revision_number":1},{"id":"f6ebf56e05f79242a133bba237fe1e4621171fb9","author":{"_account_id":28619,"name":"Dmitriy Rabotyagov","email":"noonedeadpunk@gmail.com","username":"noonedeadpunk"},"date":"2019-04-23 10:52:05.000000000","message":"Patch Set 1: Code-Review+2 Workflow+1","accounts_in_message":[],"_revision_number":1},{"id":"6d8e56063290a2c75e90da158ffecb043b894ca2","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2019-04-23 10:58:30.000000000","message":"Patch Set 1: Verified+1\n\nBuild succeeded (check pipeline).\n\n- openstack-ansible-linters http://logs.openstack.org/97/654897/1/check/openstack-ansible-linters/9ab585b/ : SUCCESS in 5m 46s\n- openstack-ansible-deploy-aio_telemetry_metal-centos-7 http://logs.openstack.org/97/654897/1/check/openstack-ansible-deploy-aio_telemetry_metal-centos-7/a97c5a0/ : SUCCESS in 1h 29m 41s\n- openstack-ansible-deploy-aio_telemetry_metal-debian-stable http://logs.openstack.org/97/654897/1/check/openstack-ansible-deploy-aio_telemetry_metal-debian-stable/c0753cc/ : SUCCESS in 1h 16m 19s\n- openstack-ansible-deploy-aio_telemetry_metal-ubuntu-bionic http://logs.openstack.org/97/654897/1/check/openstack-ansible-deploy-aio_telemetry_metal-ubuntu-bionic/3ea9692/ : SUCCESS in 1h 08m 24s\n- openstack-tox-docs http://logs.openstack.org/97/654897/1/check/openstack-tox-docs/77b9ea4/html/ : SUCCESS in 2m 45s","accounts_in_message":[],"_revision_number":1},{"id":"08b1ecf5f35da48bbc2f223a78ff97524fe44a82","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2019-04-23 10:58:41.000000000","message":"Patch Set 1: -Verified\n\nStarting gate jobs.","accounts_in_message":[],"_revision_number":1},{"id":"f4f955fcc98f4747eb4cd3be1fbff3302adf4e78","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2019-04-23 13:12:19.000000000","message":"Patch Set 1: Verified-2\n\nBuild failed (gate pipeline).  For information on how to proceed, see\nhttp://docs.openstack.org/infra/manual/developers.html#automated-testing\n\n\n- openstack-ansible-linters http://logs.openstack.org/97/654897/1/gate/openstack-ansible-linters/4572659/ : SUCCESS in 5m 55s\n- openstack-ansible-deploy-aio_telemetry_metal-centos-7 finger://ze04.openstack.org/57a09bf7edd34507852709208ce57d6e : RETRY_LIMIT in 43m 46s\n- openstack-ansible-deploy-aio_telemetry_metal-debian-stable finger://ze07.openstack.org/dd5abb71bcc640fcaf8886dcb0fe4783 : RETRY_LIMIT in 43m 53s\n- openstack-ansible-deploy-aio_telemetry_metal-ubuntu-bionic finger://ze07.openstack.org/a111fdafd77c4b7fa4f50031e92cb68f : RETRY_LIMIT in 43m 10s\n- openstack-tox-docs http://logs.openstack.org/97/654897/1/gate/openstack-tox-docs/f32c199/html/ : SUCCESS in 2m 49s","accounts_in_message":[],"_revision_number":1},{"id":"65c988b7c304bb3e5f9ba0415c2e228966c0003c","author":{"_account_id":25023,"name":"Jonathan Rosser","email":"jonathan.rosser@rd.bbc.co.uk","username":"jrosser"},"date":"2019-04-23 20:21:21.000000000","message":"Patch Set 1:\n\nrecheck","accounts_in_message":[],"_revision_number":1},{"id":"3a518d339fd4bd8844df30bca2878f4fad56acdb","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2019-04-23 22:22:48.000000000","message":"Patch Set 1: Verified+1\n\nBuild succeeded (check pipeline).\n\n- openstack-ansible-linters http://logs.openstack.org/97/654897/1/check/openstack-ansible-linters/7a9bebf/ : SUCCESS in 5m 30s\n- openstack-ansible-deploy-aio_telemetry_metal-centos-7 http://logs.openstack.org/97/654897/1/check/openstack-ansible-deploy-aio_telemetry_metal-centos-7/b9ef926/ : SUCCESS in 1h 30m 02s\n- openstack-ansible-deploy-aio_telemetry_metal-debian-stable http://logs.openstack.org/97/654897/1/check/openstack-ansible-deploy-aio_telemetry_metal-debian-stable/766fc10/ : SUCCESS in 1h 10m 42s\n- openstack-ansible-deploy-aio_telemetry_metal-ubuntu-bionic http://logs.openstack.org/97/654897/1/check/openstack-ansible-deploy-aio_telemetry_metal-ubuntu-bionic/37368a0/ : SUCCESS in 1h 44m 47s\n- openstack-tox-docs http://logs.openstack.org/97/654897/1/check/openstack-tox-docs/24ba48b/html/ : SUCCESS in 3m 48s","accounts_in_message":[],"_revision_number":1},{"id":"b995be28d0a6b2c09d3047617e7940d80ec4fef1","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2019-04-23 22:23:01.000000000","message":"Patch Set 1: -Verified\n\nStarting gate jobs.","accounts_in_message":[],"_revision_number":1},{"id":"45d2614d5bb23c4a6d8e3392b8e4d3463b5a82fc","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2019-04-24 00:11:59.000000000","message":"Patch Set 1: Verified+2\n\nBuild succeeded (gate pipeline).\n\n- openstack-ansible-linters http://logs.openstack.org/97/654897/1/gate/openstack-ansible-linters/65fb23f/ : SUCCESS in 5m 43s\n- openstack-ansible-deploy-aio_telemetry_metal-centos-7 http://logs.openstack.org/97/654897/1/gate/openstack-ansible-deploy-aio_telemetry_metal-centos-7/667b01b/ : SUCCESS in 1h 42m 58s\n- openstack-ansible-deploy-aio_telemetry_metal-debian-stable http://logs.openstack.org/97/654897/1/gate/openstack-ansible-deploy-aio_telemetry_metal-debian-stable/87a4811/ : SUCCESS in 56m 25s\n- openstack-ansible-deploy-aio_telemetry_metal-ubuntu-bionic http://logs.openstack.org/97/654897/1/gate/openstack-ansible-deploy-aio_telemetry_metal-ubuntu-bionic/29fe69c/ : SUCCESS in 1h 22m 16s\n- openstack-tox-docs http://logs.openstack.org/97/654897/1/gate/openstack-tox-docs/9c2950e/html/ : SUCCESS in 2m 39s","accounts_in_message":[],"_revision_number":1},{"id":"46b4cccebe4f6e8b675c3274d5048eeec62093a8","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2019-04-24 00:12:00.000000000","message":"Change has been successfully merged by Zuul","accounts_in_message":[],"_revision_number":1}],"current_revision_number":1,"current_revision":"0de8b3d69010ad1a86ace9dce9c6edf96ef6848c","revisions":{"0de8b3d69010ad1a86ace9dce9c6edf96ef6848c":{"kind":"REWORK","_number":1,"created":"2019-04-23 09:23:37.000000000","uploader":{"_account_id":11131,"name":"OpenStack Proposal Bot","username":"proposal-bot","tags":["SERVICE_USER"]},"ref":"refs/changes/97/654897/1","fetch":{"anonymous http":{"url":"https://review.opendev.org/openstack/openstack-ansible-os_aodh","ref":"refs/changes/97/654897/1","commands":{"Checkout":"git fetch https://review.opendev.org/openstack/openstack-ansible-os_aodh refs/changes/97/654897/1 \u0026\u0026 git checkout FETCH_HEAD","Cherry Pick":"git fetch https://review.opendev.org/openstack/openstack-ansible-os_aodh refs/changes/97/654897/1 \u0026\u0026 git cherry-pick FETCH_HEAD","Format Patch":"git fetch https://review.opendev.org/openstack/openstack-ansible-os_aodh refs/changes/97/654897/1 \u0026\u0026 git format-patch -1 --stdout FETCH_HEAD","Pull":"git pull https://review.opendev.org/openstack/openstack-ansible-os_aodh refs/changes/97/654897/1"}}},"commit":{"parents":[{"commit":"c9b029ad7f4804dc03cc0c62e27ed4985c85a014","subject":"Updated from OpenStack Ansible Tests","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/openstack-ansible-os_aodh/commit/c9b029ad7f4804dc03cc0c62e27ed4985c85a014"}]}],"author":{"name":"OpenStack Proposal Bot","email":"openstack-infra@lists.openstack.org","date":"2019-04-23 09:23:36.000000000","tz":0},"committer":{"name":"OpenStack Proposal Bot","email":"openstack-infra@lists.openstack.org","date":"2019-04-23 09:23:36.000000000","tz":0},"subject":"Updated from OpenStack Ansible Tests","message":"Updated from OpenStack Ansible Tests\n\nChange-Id: I7429f11f46a893ab93982aad81bc2f422f6a8b66\n","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/openstack-ansible-os_aodh/commit/0de8b3d69010ad1a86ace9dce9c6edf96ef6848c"}],"resolve_conflicts_web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/openstack-ansible-os_aodh/commit/0de8b3d69010ad1a86ace9dce9c6edf96ef6848c"}]},"branch":"refs/heads/stable/stein"}},"requirements":[],"submit_records":[],"submit_requirements":[]}
