)]}'
{"id":"openstack%2Fopenstack-ansible-os_aodh~655503","triplet_id":"openstack%2Fopenstack-ansible-os_aodh~master~I819feb9489902329adccde46883dc5c802259df4","project":"openstack/openstack-ansible-os_aodh","branch":"master","topic":"bug/1826250","hashtags":[],"change_id":"I819feb9489902329adccde46883dc5c802259df4","subject":"Adding aodh_service_timeout","status":"ABANDONED","created":"2019-04-24 17:19:34.000000000","updated":"2019-07-30 19:46:03.000000000","total_comment_count":0,"unresolved_comment_count":0,"has_review_started":true,"meta_rev_id":"e8ef28faeeab24e88d8daee28586a0ccd5c54d59","_number":655503,"virtual_id_number":655503,"owner":{"_account_id":14552,"name":"Bjoern Teipel","email":"bjoern.teipel@rackspace.com","username":"bjoern"},"actions":{},"labels":{"Verified":{"recommended":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"all":[{"_account_id":1004,"name":"Mohammed Naser","email":"mnaser@vexxhost.com","username":"mnaser"},{"date":"2019-04-26 16:36:30.000000000","_account_id":17068,"name":"Jean-Philippe Evrard","email":"openstack@a.spamming.party","username":"evrardjp"},{"value":1,"date":"2019-04-25 01:09:22.000000000","permitted_voting_range":{"min":-2,"max":2},"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]}],"values":{"-2":"Fails","-1":"Doesn\u0027t seem to work"," 0":"No score","+1":"Works for me","+2":"Verified"},"description":"","value":1,"default_value":0,"optional":true},"Code-Review":{"disliked":{"_account_id":1004,"name":"Mohammed Naser","email":"mnaser@vexxhost.com","username":"mnaser"},"all":[{"value":-1,"date":"2019-04-26 17:34:59.000000000","permitted_voting_range":{"min":-2,"max":2},"_account_id":1004,"name":"Mohammed Naser","email":"mnaser@vexxhost.com","username":"mnaser"},{"value":0,"permitted_voting_range":{"min":-1,"max":1},"_account_id":17068,"name":"Jean-Philippe Evrard","email":"openstack@a.spamming.party","username":"evrardjp"},{"value":0,"permitted_voting_range":{"min":-1,"max":1},"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]}],"values":{"-2":"Do not merge","-1":"This patch needs further work before it can be merged"," 0":"No score","+1":"Looks good to me, but someone else must approve","+2":"Looks good to me (core reviewer)"},"description":"","value":-1,"default_value":0,"optional":true},"Workflow":{"all":[{"value":0,"permitted_voting_range":{"min":-1,"max":1},"_account_id":1004,"name":"Mohammed Naser","email":"mnaser@vexxhost.com","username":"mnaser"},{"_account_id":17068,"name":"Jean-Philippe Evrard","email":"openstack@a.spamming.party","username":"evrardjp"},{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]}],"values":{"-1":"Work in progress"," 0":"Ready for reviews","+1":"Approved"},"description":"","default_value":0,"optional":true},"Backport-Candidate":{"all":[{"value":0,"permitted_voting_range":{"min":-1,"max":1},"_account_id":1004,"name":"Mohammed Naser","email":"mnaser@vexxhost.com","username":"mnaser"},{"_account_id":17068,"name":"Jean-Philippe Evrard","email":"openstack@a.spamming.party","username":"evrardjp"},{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]}],"values":{"-1":"Not A Backport Candidate"," 0":"Backport Review Needed","+1":"Proposed Backport"},"description":"","default_value":0,"optional":true}},"removable_reviewers":[],"reviewers":{"REVIEWER":[{"_account_id":1004,"name":"Mohammed Naser","email":"mnaser@vexxhost.com","username":"mnaser"},{"_account_id":17068,"name":"Jean-Philippe Evrard","email":"openstack@a.spamming.party","username":"evrardjp"},{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]}]},"pending_reviewers":{},"reviewer_updates":[{"updated":"2019-04-25 01:09:22.000000000","updated_by":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"reviewer":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"state":"REVIEWER"},{"updated":"2019-04-26 16:36:30.000000000","updated_by":{"_account_id":17068,"name":"Jean-Philippe Evrard","email":"openstack@a.spamming.party","username":"evrardjp"},"reviewer":{"_account_id":17068,"name":"Jean-Philippe Evrard","email":"openstack@a.spamming.party","username":"evrardjp"},"state":"REVIEWER"},{"updated":"2019-04-26 17:34:59.000000000","updated_by":{"_account_id":1004,"name":"Mohammed Naser","email":"mnaser@vexxhost.com","username":"mnaser"},"reviewer":{"_account_id":1004,"name":"Mohammed Naser","email":"mnaser@vexxhost.com","username":"mnaser"},"state":"REVIEWER"}],"messages":[{"id":"80f972ef4ed7dd4ffeef71f93a63cf290c810f21","author":{"_account_id":14552,"name":"Bjoern Teipel","email":"bjoern.teipel@rackspace.com","username":"bjoern"},"date":"2019-04-24 17:19:34.000000000","message":"Uploaded patch set 1.","accounts_in_message":[],"_revision_number":1},{"id":"8cfab0cca9bcfd1a86dbec77ea298536742b6683","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2019-04-24 19:07:14.000000000","message":"Patch Set 1: Verified+1\n\nBuild succeeded (check pipeline).\n\n- openstack-ansible-linters http://logs.openstack.org/03/655503/1/check/openstack-ansible-linters/7aa0033/ : SUCCESS in 4m 21s\n- openstack-ansible-deploy-aio_telemetry_metal-centos-7 http://logs.openstack.org/03/655503/1/check/openstack-ansible-deploy-aio_telemetry_metal-centos-7/3115288/ : SUCCESS in 1h 38m 29s\n- openstack-ansible-deploy-aio_telemetry_metal-debian-stable http://logs.openstack.org/03/655503/1/check/openstack-ansible-deploy-aio_telemetry_metal-debian-stable/e0d2a54/ : SUCCESS in 1h 20m 17s\n- openstack-ansible-deploy-aio_telemetry_metal-ubuntu-bionic http://logs.openstack.org/03/655503/1/check/openstack-ansible-deploy-aio_telemetry_metal-ubuntu-bionic/7fd9cbd/ : SUCCESS in 1h 06m 27s\n- openstack-tox-docs http://logs.openstack.org/03/655503/1/check/openstack-tox-docs/7ce8a9e/html/ : SUCCESS in 2m 36s\n- build-openstack-releasenotes http://logs.openstack.org/03/655503/1/check/build-openstack-releasenotes/71af6b1/html/ : SUCCESS in 3m 31s","accounts_in_message":[],"_revision_number":1},{"id":"9759427665b99444b376a0cc4901cfadaf9fbae8","author":{"_account_id":14552,"name":"Bjoern Teipel","email":"bjoern.teipel@rackspace.com","username":"bjoern"},"date":"2019-04-24 20:06:00.000000000","message":"Uploaded patch set 2.","accounts_in_message":[],"_revision_number":2},{"id":"42c0bb5d0f76f5bf083640175818352482e8f846","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2019-04-25 01:09:22.000000000","message":"Patch Set 2: Verified+1\n\nBuild succeeded (check pipeline).\n\n- openstack-ansible-linters http://logs.openstack.org/03/655503/2/check/openstack-ansible-linters/10509f2/ : SUCCESS in 5m 54s\n- openstack-ansible-deploy-aio_telemetry_metal-centos-7 http://logs.openstack.org/03/655503/2/check/openstack-ansible-deploy-aio_telemetry_metal-centos-7/130844e/ : SUCCESS in 1h 30m 20s\n- openstack-ansible-deploy-aio_telemetry_metal-debian-stable http://logs.openstack.org/03/655503/2/check/openstack-ansible-deploy-aio_telemetry_metal-debian-stable/d407d1d/ : SUCCESS in 1h 10m 47s\n- openstack-ansible-deploy-aio_telemetry_metal-ubuntu-bionic http://logs.openstack.org/03/655503/2/check/openstack-ansible-deploy-aio_telemetry_metal-ubuntu-bionic/da617a2/ : SUCCESS in 1h 06m 19s\n- openstack-tox-docs http://logs.openstack.org/03/655503/2/check/openstack-tox-docs/dbc885d/html/ : SUCCESS in 2m 35s\n- build-openstack-releasenotes http://logs.openstack.org/03/655503/2/check/build-openstack-releasenotes/4d60f2f/html/ : SUCCESS in 3m 04s","accounts_in_message":[],"_revision_number":2},{"id":"7efdaefe32916c2db504af960306c806b80a3265","author":{"_account_id":17068,"name":"Jean-Philippe Evrard","email":"openstack@a.spamming.party","username":"evrardjp"},"date":"2019-04-26 16:36:30.000000000","message":"Patch Set 2:\n\nSee comment in https://review.opendev.org/#/c/655526/2","accounts_in_message":[],"_revision_number":2},{"id":"5e28e2441e21d40c9c5c8756a39bb2ab671c6ebb","author":{"_account_id":1004,"name":"Mohammed Naser","email":"mnaser@vexxhost.com","username":"mnaser"},"date":"2019-04-26 17:34:59.000000000","message":"Patch Set 2: Code-Review-1\n\nplease review feedback on https://review.opendev.org/#/c/655559/1","accounts_in_message":[],"_revision_number":2},{"id":"3efe77c558724c7f32ecdaf1574993843416f1c8","author":{"_account_id":1004,"name":"Mohammed Naser","email":"mnaser@vexxhost.com","username":"mnaser"},"date":"2019-07-30 19:46:03.000000000","message":"Abandoned\n\nno follow up patch and adds an unnecessary variable which can just be used by an override","accounts_in_message":[],"_revision_number":2}],"current_revision_number":2,"current_revision":"8a45fb684600eb35079570ec98cc1fa841b62141","revisions":{"f8b4f24280486468c56bb1820499821c9806be7c":{"kind":"REWORK","_number":1,"created":"2019-04-24 17:19:34.000000000","uploader":{"_account_id":14552,"name":"Bjoern Teipel","email":"bjoern.teipel@rackspace.com","username":"bjoern"},"ref":"refs/changes/03/655503/1","fetch":{"anonymous http":{"url":"https://review.opendev.org/openstack/openstack-ansible-os_aodh","ref":"refs/changes/03/655503/1","commands":{"Checkout":"git fetch https://review.opendev.org/openstack/openstack-ansible-os_aodh refs/changes/03/655503/1 \u0026\u0026 git checkout FETCH_HEAD","Cherry Pick":"git fetch https://review.opendev.org/openstack/openstack-ansible-os_aodh refs/changes/03/655503/1 \u0026\u0026 git cherry-pick FETCH_HEAD","Format Patch":"git fetch https://review.opendev.org/openstack/openstack-ansible-os_aodh refs/changes/03/655503/1 \u0026\u0026 git format-patch -1 --stdout FETCH_HEAD","Pull":"git pull https://review.opendev.org/openstack/openstack-ansible-os_aodh refs/changes/03/655503/1"}}},"commit":{"parents":[{"commit":"58b01a8599c2d213a72eb3a51e843d71b873a502","subject":"Updated from OpenStack Ansible Tests","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/openstack-ansible-os_aodh/commit/58b01a8599c2d213a72eb3a51e843d71b873a502"}]}],"author":{"name":"Bjoern Teipel","email":"bjoern.teipel@rackspace.com","date":"2019-04-24 17:18:51.000000000","tz":-300},"committer":{"name":"Bjoern Teipel","email":"bjoern.teipel@rackspace.com","date":"2019-04-24 17:18:51.000000000","tz":-300},"subject":"Adding aodh_service_timeout","message":"Adding aodh_service_timeout\n\nThe override `aodh_service_timeout` is introduced to allow\nmultiple service components to use the same frontend/backend\ntimeouts. The default value is 60 (seconds).\n\nChange-Id: I819feb9489902329adccde46883dc5c802259df4\nCloses-Bug: 1826250\n","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/openstack-ansible-os_aodh/commit/f8b4f24280486468c56bb1820499821c9806be7c"}],"resolve_conflicts_web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/openstack-ansible-os_aodh/commit/f8b4f24280486468c56bb1820499821c9806be7c"}]},"branch":"refs/heads/master"},"8a45fb684600eb35079570ec98cc1fa841b62141":{"kind":"REWORK","_number":2,"created":"2019-04-24 20:06:00.000000000","uploader":{"_account_id":14552,"name":"Bjoern Teipel","email":"bjoern.teipel@rackspace.com","username":"bjoern"},"ref":"refs/changes/03/655503/2","fetch":{"anonymous http":{"url":"https://review.opendev.org/openstack/openstack-ansible-os_aodh","ref":"refs/changes/03/655503/2","commands":{"Checkout":"git fetch https://review.opendev.org/openstack/openstack-ansible-os_aodh refs/changes/03/655503/2 \u0026\u0026 git checkout FETCH_HEAD","Cherry Pick":"git fetch https://review.opendev.org/openstack/openstack-ansible-os_aodh refs/changes/03/655503/2 \u0026\u0026 git cherry-pick FETCH_HEAD","Format Patch":"git fetch https://review.opendev.org/openstack/openstack-ansible-os_aodh refs/changes/03/655503/2 \u0026\u0026 git format-patch -1 --stdout FETCH_HEAD","Pull":"git pull https://review.opendev.org/openstack/openstack-ansible-os_aodh refs/changes/03/655503/2"}}},"commit":{"parents":[{"commit":"58b01a8599c2d213a72eb3a51e843d71b873a502","subject":"Updated from OpenStack Ansible Tests","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/openstack-ansible-os_aodh/commit/58b01a8599c2d213a72eb3a51e843d71b873a502"}]}],"author":{"name":"Bjoern Teipel","email":"bjoern.teipel@rackspace.com","date":"2019-04-24 17:18:51.000000000","tz":-300},"committer":{"name":"Bjoern Teipel","email":"bjoern.teipel@rackspace.com","date":"2019-04-24 20:05:55.000000000","tz":-300},"subject":"Adding aodh_service_timeout","message":"Adding aodh_service_timeout\n\nThe override `aodh_service_timeout` is introduced to allow\nmultiple service components to use the same frontend/backend\ntimeouts. The default value is 60 (seconds).\n\nChange-Id: I819feb9489902329adccde46883dc5c802259df4\nCloses-Bug: 1826250\n","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/openstack-ansible-os_aodh/commit/8a45fb684600eb35079570ec98cc1fa841b62141"}],"resolve_conflicts_web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/openstack-ansible-os_aodh/commit/8a45fb684600eb35079570ec98cc1fa841b62141"}]},"branch":"refs/heads/master"}},"requirements":[],"submit_records":[],"submit_requirements":[]}
