)]}'
{"id":"openstack%2Fopenstack-ansible-os_aodh~657243","triplet_id":"openstack%2Fopenstack-ansible-os_aodh~master~I1231b4d0d0b0865eaf88d42ad3a4c69aa1c2a961","project":"openstack/openstack-ansible-os_aodh","branch":"master","hashtags":[],"change_id":"I1231b4d0d0b0865eaf88d42ad3a4c69aa1c2a961","subject":"Replace git.openstack.org URLs with opendev.org URLs","status":"MERGED","created":"2019-05-05 12:04:18.000000000","updated":"2019-07-17 17:26:02.000000000","submitted":"2019-07-17 17:26:02.000000000","submitter":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"total_comment_count":2,"unresolved_comment_count":0,"has_review_started":true,"submission_id":"657243-1563384362254-e47bf6dd","meta_rev_id":"1aa79236ad70b7b1e4a2e94ed0afa1127de7e9da","_number":657243,"virtual_id_number":657243,"owner":{"_account_id":17130,"name":"melissaml","email":"ma.lei@99cloud.net","username":"malei"},"actions":{},"labels":{"Verified":{"approved":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"all":[{"value":0,"_account_id":18279,"name":"Jonathan Herlin","email":"jonte@jherlin.se","username":"Jonher937"},{"value":0,"_account_id":28935,"name":"98k","email":"18552437190@163.com","username":"98k"},{"value":0,"_account_id":28619,"name":"Dmitriy Rabotyagov","email":"noonedeadpunk@gmail.com","username":"noonedeadpunk"},{"value":2,"date":"2019-07-17 17:26:02.000000000","permitted_voting_range":{"min":2,"max":2},"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},{"value":0,"_account_id":25023,"name":"Jonathan Rosser","email":"jonathan.rosser@rd.bbc.co.uk","username":"jrosser"},{"value":0,"_account_id":28543,"name":"shupeng","email":"15050873171@163.com","username":"Vieri"}],"values":{"-2":"Fails","-1":"Doesn\u0027t seem to work"," 0":"No score","+1":"Works for me","+2":"Verified"},"description":"","default_value":0,"optional":true},"Code-Review":{"approved":{"_account_id":28619,"name":"Dmitriy Rabotyagov","email":"noonedeadpunk@gmail.com","username":"noonedeadpunk"},"recommended":{"_account_id":18279,"name":"Jonathan Herlin","email":"jonte@jherlin.se","username":"Jonher937"},"all":[{"value":1,"date":"2019-07-17 06:33:43.000000000","permitted_voting_range":{"min":1,"max":1},"_account_id":18279,"name":"Jonathan Herlin","email":"jonte@jherlin.se","username":"Jonher937"},{"value":0,"_account_id":28935,"name":"98k","email":"18552437190@163.com","username":"98k"},{"value":2,"date":"2019-07-17 05:37:57.000000000","permitted_voting_range":{"min":2,"max":2},"_account_id":28619,"name":"Dmitriy Rabotyagov","email":"noonedeadpunk@gmail.com","username":"noonedeadpunk"},{"value":0,"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},{"value":2,"date":"2019-07-17 15:21:12.000000000","permitted_voting_range":{"min":2,"max":2},"_account_id":25023,"name":"Jonathan Rosser","email":"jonathan.rosser@rd.bbc.co.uk","username":"jrosser"},{"value":0,"_account_id":28543,"name":"shupeng","email":"15050873171@163.com","username":"Vieri"}],"values":{"-2":"Do not merge","-1":"This patch needs further work before it can be merged"," 0":"No score","+1":"Looks good to me, but someone else must approve","+2":"Looks good to me (core reviewer)"},"description":"","value":1,"default_value":0,"optional":true},"Workflow":{"approved":{"_account_id":25023,"name":"Jonathan Rosser","email":"jonathan.rosser@rd.bbc.co.uk","username":"jrosser"},"all":[{"value":0,"_account_id":18279,"name":"Jonathan Herlin","email":"jonte@jherlin.se","username":"Jonher937"},{"value":0,"_account_id":28935,"name":"98k","email":"18552437190@163.com","username":"98k"},{"value":0,"_account_id":28619,"name":"Dmitriy Rabotyagov","email":"noonedeadpunk@gmail.com","username":"noonedeadpunk"},{"value":0,"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},{"value":1,"date":"2019-07-17 15:21:12.000000000","permitted_voting_range":{"min":1,"max":1},"_account_id":25023,"name":"Jonathan Rosser","email":"jonathan.rosser@rd.bbc.co.uk","username":"jrosser"},{"value":0,"_account_id":28543,"name":"shupeng","email":"15050873171@163.com","username":"Vieri"}],"values":{"-1":"Work in progress"," 0":"Ready for reviews","+1":"Approved"},"description":"","default_value":0,"optional":true},"Backport-Candidate":{"all":[{"value":0,"_account_id":18279,"name":"Jonathan Herlin","email":"jonte@jherlin.se","username":"Jonher937"},{"value":0,"_account_id":28935,"name":"98k","email":"18552437190@163.com","username":"98k"},{"value":0,"_account_id":28619,"name":"Dmitriy Rabotyagov","email":"noonedeadpunk@gmail.com","username":"noonedeadpunk"},{"value":0,"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},{"value":0,"_account_id":25023,"name":"Jonathan Rosser","email":"jonathan.rosser@rd.bbc.co.uk","username":"jrosser"},{"value":0,"_account_id":28543,"name":"shupeng","email":"15050873171@163.com","username":"Vieri"}],"values":{"-1":"Not A Backport Candidate"," 0":"Backport Review Needed","+1":"Proposed Backport"},"description":"","default_value":0,"optional":true}},"removable_reviewers":[],"reviewers":{"REVIEWER":[{"_account_id":18279,"name":"Jonathan Herlin","email":"jonte@jherlin.se","username":"Jonher937"},{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},{"_account_id":25023,"name":"Jonathan Rosser","email":"jonathan.rosser@rd.bbc.co.uk","username":"jrosser"},{"_account_id":28543,"name":"shupeng","email":"15050873171@163.com","username":"Vieri"},{"_account_id":28619,"name":"Dmitriy Rabotyagov","email":"noonedeadpunk@gmail.com","username":"noonedeadpunk"},{"_account_id":28935,"name":"98k","email":"18552437190@163.com","username":"98k"}]},"pending_reviewers":{},"reviewer_updates":[{"updated":"2019-07-14 12:32:40.000000000","updated_by":{"_account_id":28935,"name":"98k","email":"18552437190@163.com","username":"98k"},"reviewer":{"_account_id":28935,"name":"98k","email":"18552437190@163.com","username":"98k"},"state":"REVIEWER"},{"updated":"2019-07-14 13:39:40.000000000","updated_by":{"_account_id":28543,"name":"shupeng","email":"15050873171@163.com","username":"Vieri"},"reviewer":{"_account_id":28543,"name":"shupeng","email":"15050873171@163.com","username":"Vieri"},"state":"REVIEWER"},{"updated":"2019-07-17 05:37:57.000000000","updated_by":{"_account_id":28619,"name":"Dmitriy Rabotyagov","email":"noonedeadpunk@gmail.com","username":"noonedeadpunk"},"reviewer":{"_account_id":28619,"name":"Dmitriy Rabotyagov","email":"noonedeadpunk@gmail.com","username":"noonedeadpunk"},"state":"REVIEWER"},{"updated":"2019-07-17 06:33:43.000000000","updated_by":{"_account_id":18279,"name":"Jonathan Herlin","email":"jonte@jherlin.se","username":"Jonher937"},"reviewer":{"_account_id":18279,"name":"Jonathan Herlin","email":"jonte@jherlin.se","username":"Jonher937"},"state":"REVIEWER"},{"updated":"2019-07-17 15:21:12.000000000","updated_by":{"_account_id":25023,"name":"Jonathan Rosser","email":"jonathan.rosser@rd.bbc.co.uk","username":"jrosser"},"reviewer":{"_account_id":25023,"name":"Jonathan Rosser","email":"jonathan.rosser@rd.bbc.co.uk","username":"jrosser"},"state":"REVIEWER"},{"updated":"2019-07-17 17:26:02.000000000","updated_by":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"reviewer":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"state":"REVIEWER"}],"messages":[{"id":"6d7e0bd9af8542aa25651913367669831b9c4d91","author":{"_account_id":17130,"name":"melissaml","email":"ma.lei@99cloud.net","username":"malei"},"date":"2019-05-05 12:04:18.000000000","message":"Uploaded patch set 1.","accounts_in_message":[],"_revision_number":1},{"id":"946e5b7d31d2d7ce14045d8aefca2a16b8abe417","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2019-05-05 13:45:11.000000000","message":"Patch Set 1: Verified+1\n\nBuild succeeded (check pipeline).\n\n- openstack-ansible-linters http://logs.openstack.org/43/657243/1/check/openstack-ansible-linters/538f312/ : SUCCESS in 5m 10s\n- openstack-ansible-deploy-aio_telemetry_metal-centos-7 http://logs.openstack.org/43/657243/1/check/openstack-ansible-deploy-aio_telemetry_metal-centos-7/bffdac6/ : SUCCESS in 1h 24m 03s\n- openstack-ansible-deploy-aio_telemetry_metal-debian-stable http://logs.openstack.org/43/657243/1/check/openstack-ansible-deploy-aio_telemetry_metal-debian-stable/66c0ade/ : SUCCESS in 1h 08m 11s\n- openstack-ansible-deploy-aio_telemetry_metal-ubuntu-bionic http://logs.openstack.org/43/657243/1/check/openstack-ansible-deploy-aio_telemetry_metal-ubuntu-bionic/598e1f6/ : SUCCESS in 1h 09m 52s\n- openstack-tox-docs http://logs.openstack.org/43/657243/1/check/openstack-tox-docs/2554fb2/html/ : SUCCESS in 2m 45s","accounts_in_message":[],"_revision_number":1},{"id":"2d24bc15da764cd62ca5a3a4362a360ec4b984ec","author":{"_account_id":17130,"name":"melissaml","email":"ma.lei@99cloud.net","username":"malei"},"date":"2019-05-06 10:48:47.000000000","message":"Uploaded patch set 2.","accounts_in_message":[],"_revision_number":2},{"id":"50ef2712cb663c191ee7ec2261b4006040f00b83","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2019-05-06 12:14:06.000000000","message":"Patch Set 2: Verified-1\n\nBuild failed (check pipeline).  For information on how to proceed, see\nhttp://docs.openstack.org/infra/manual/developers.html#automated-testing\n\n\n- openstack-ansible-linters http://logs.openstack.org/43/657243/2/check/openstack-ansible-linters/4a131eb/ : FAILURE in 4m 18s\n- openstack-ansible-deploy-aio_telemetry_metal-centos-7 http://logs.openstack.org/43/657243/2/check/openstack-ansible-deploy-aio_telemetry_metal-centos-7/486c3f1/ : SUCCESS in 1h 19m 03s\n- openstack-ansible-deploy-aio_telemetry_metal-debian-stable http://logs.openstack.org/43/657243/2/check/openstack-ansible-deploy-aio_telemetry_metal-debian-stable/542f428/ : SUCCESS in 1h 20m 12s\n- openstack-ansible-deploy-aio_telemetry_metal-ubuntu-bionic http://logs.openstack.org/43/657243/2/check/openstack-ansible-deploy-aio_telemetry_metal-ubuntu-bionic/ee051e9/ : SUCCESS in 1h 20m 49s\n- openstack-tox-docs http://logs.openstack.org/43/657243/2/check/openstack-tox-docs/dff4ef7/ : POST_FAILURE in 2m 21s\n- build-openstack-releasenotes http://logs.openstack.org/43/657243/2/check/build-openstack-releasenotes/e3c970f/html/ : SUCCESS in 3m 07s","accounts_in_message":[],"_revision_number":2},{"id":"a8e6e51e4d660aec65c49ab9060642cc28ccf52e","author":{"_account_id":17130,"name":"melissaml","email":"ma.lei@99cloud.net","username":"malei"},"date":"2019-07-10 14:38:03.000000000","message":"Uploaded patch set 3.","accounts_in_message":[],"_revision_number":3},{"id":"c864c39f71f9fc81591dd39cebaeb6edfa79544d","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2019-07-10 16:43:13.000000000","message":"Patch Set 3: Verified-1\n\nBuild failed (check pipeline).  For information on how to proceed, see\nhttp://docs.openstack.org/infra/manual/developers.html#automated-testing\n\n\n- openstack-ansible-linters http://logs.openstack.org/43/657243/3/check/openstack-ansible-linters/c504d3e/ : FAILURE in 2m 39s\n- openstack-ansible-deploy-aio_metal-debian-stable http://logs.openstack.org/43/657243/3/check/openstack-ansible-deploy-aio_metal-debian-stable/d20c1df/ : SUCCESS in 1h 41m 04s\n- openstack-ansible-deploy-aio_metal-centos-7 http://logs.openstack.org/43/657243/3/check/openstack-ansible-deploy-aio_metal-centos-7/9f16fc2/ : SUCCESS in 1h 43m 56s\n- openstack-ansible-deploy-aio_metal-opensuse-15 http://logs.openstack.org/43/657243/3/check/openstack-ansible-deploy-aio_metal-opensuse-15/18a33f2/ : SUCCESS in 1h 47m 42s\n- openstack-ansible-deploy-aio_metal-ubuntu-bionic http://logs.openstack.org/43/657243/3/check/openstack-ansible-deploy-aio_metal-ubuntu-bionic/5edec92/ : SUCCESS in 1h 13m 46s\n- openstack-ansible-deploy-aio_distro_metal-centos-7 http://logs.openstack.org/43/657243/3/check/openstack-ansible-deploy-aio_distro_metal-centos-7/87ec9b4/ : SUCCESS in 1h 16m 11s\n- openstack-ansible-deploy-aio_distro_metal-opensuse-15 http://logs.openstack.org/43/657243/3/check/openstack-ansible-deploy-aio_distro_metal-opensuse-15/bd1efc6/ : FAILURE in 30m 12s (non-voting)\n- openstack-ansible-deploy-aio_telemetry_metal-centos-7 http://logs.openstack.org/43/657243/3/check/openstack-ansible-deploy-aio_telemetry_metal-centos-7/2a773f5/ : SUCCESS in 1h 53m 30s\n- openstack-ansible-deploy-aio_telemetry_metal-debian-stable http://logs.openstack.org/43/657243/3/check/openstack-ansible-deploy-aio_telemetry_metal-debian-stable/1c50450/ : SUCCESS in 1h 14m 06s\n- openstack-ansible-deploy-aio_telemetry_metal-ubuntu-bionic http://logs.openstack.org/43/657243/3/check/openstack-ansible-deploy-aio_telemetry_metal-ubuntu-bionic/7868256/ : SUCCESS in 1h 09m 12s\n- openstack-tox-docs http://logs.openstack.org/43/657243/3/check/openstack-tox-docs/555f6c9/ : POST_FAILURE in 3m 05s\n- build-openstack-releasenotes http://logs.openstack.org/43/657243/3/check/build-openstack-releasenotes/30ef476/html/ : SUCCESS in 3m 57s","accounts_in_message":[],"_revision_number":3},{"id":"4ef3ca2f5b6b54a4738397dca61b08c15d6150ac","author":{"_account_id":18279,"name":"Jonathan Herlin","email":"jonte@jherlin.se","username":"Jonher937"},"date":"2019-07-13 19:29:44.000000000","message":"Patch Set 3: Code-Review-1\n\n(1 comment)","accounts_in_message":[],"_revision_number":3},{"id":"03c8985a05c261b9859191a3bd57fd3161eca519","author":{"_account_id":17130,"name":"melissaml","email":"ma.lei@99cloud.net","username":"malei"},"date":"2019-07-13 19:37:29.000000000","message":"Uploaded patch set 4.","accounts_in_message":[],"_revision_number":4},{"id":"01ffd28f9c714ac93df8da2c82509852d3cceb70","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2019-07-13 21:28:08.000000000","message":"Patch Set 4: Verified+1\n\nBuild succeeded (check pipeline).\n\n- openstack-ansible-linters http://logs.openstack.org/43/657243/4/check/openstack-ansible-linters/0719011/ : SUCCESS in 4m 14s\n- openstack-ansible-deploy-aio_metal-debian-stable http://logs.openstack.org/43/657243/4/check/openstack-ansible-deploy-aio_metal-debian-stable/ac35ec0/ : SUCCESS in 1h 16m 14s\n- openstack-ansible-deploy-aio_metal-centos-7 http://logs.openstack.org/43/657243/4/check/openstack-ansible-deploy-aio_metal-centos-7/7badacc/ : SUCCESS in 1h 40m 17s\n- openstack-ansible-deploy-aio_metal-opensuse-15 http://logs.openstack.org/43/657243/4/check/openstack-ansible-deploy-aio_metal-opensuse-15/06edc82/ : SUCCESS in 1h 23m 53s\n- openstack-ansible-deploy-aio_metal-ubuntu-bionic http://logs.openstack.org/43/657243/4/check/openstack-ansible-deploy-aio_metal-ubuntu-bionic/05fbcf6/ : SUCCESS in 1h 02m 59s\n- openstack-ansible-deploy-aio_distro_metal-centos-7 http://logs.openstack.org/43/657243/4/check/openstack-ansible-deploy-aio_distro_metal-centos-7/61f6c17/ : SUCCESS in 1h 05m 35s\n- openstack-ansible-deploy-aio_distro_metal-opensuse-15 http://logs.openstack.org/43/657243/4/check/openstack-ansible-deploy-aio_distro_metal-opensuse-15/24f4df0/ : FAILURE in 51m 24s (non-voting)\n- openstack-ansible-deploy-aio_telemetry_metal-centos-7 http://logs.openstack.org/43/657243/4/check/openstack-ansible-deploy-aio_telemetry_metal-centos-7/a210282/ : SUCCESS in 1h 45m 40s\n- openstack-ansible-deploy-aio_telemetry_metal-debian-stable http://logs.openstack.org/43/657243/4/check/openstack-ansible-deploy-aio_telemetry_metal-debian-stable/b950f84/ : SUCCESS in 1h 16m 14s\n- openstack-ansible-deploy-aio_telemetry_metal-ubuntu-bionic http://logs.openstack.org/43/657243/4/check/openstack-ansible-deploy-aio_telemetry_metal-ubuntu-bionic/aa3e51d/ : SUCCESS in 1h 09m 24s\n- openstack-tox-docs http://logs.openstack.org/43/657243/4/check/openstack-tox-docs/b46681d/html/ : SUCCESS in 2m 34s\n- build-openstack-releasenotes http://logs.openstack.org/43/657243/4/check/build-openstack-releasenotes/8467b0b/html/ : SUCCESS in 3m 21s","accounts_in_message":[],"_revision_number":4},{"id":"b230546d8679d737a514f554d0ed444288668a07","author":{"_account_id":18279,"name":"Jonathan Herlin","email":"jonte@jherlin.se","username":"Jonher937"},"date":"2019-07-13 21:29:52.000000000","message":"Patch Set 4: Code-Review+1","accounts_in_message":[],"_revision_number":4},{"id":"f330288e69dad5fb314a6084c7ab26602c6a80e5","author":{"_account_id":28935,"name":"98k","email":"18552437190@163.com","username":"98k"},"date":"2019-07-14 12:32:40.000000000","message":"Patch Set 4: Code-Review+1","accounts_in_message":[],"_revision_number":4},{"id":"fb86f0f3f5a462c77c9a7b7e0a08e01aa0cbe99e","author":{"_account_id":28543,"name":"shupeng","email":"15050873171@163.com","username":"Vieri"},"date":"2019-07-14 13:39:40.000000000","message":"Patch Set 4: Code-Review+1","accounts_in_message":[],"_revision_number":4},{"id":"25ed6de69a38b2cc06bf0059267808129c0ae43f","author":{"_account_id":28619,"name":"Dmitriy Rabotyagov","email":"noonedeadpunk@gmail.com","username":"noonedeadpunk"},"date":"2019-07-16 14:21:35.000000000","message":"Patch Set 4:\n\n(1 comment)","accounts_in_message":[],"_revision_number":4},{"id":"cb3c0cf6cc0521321d87be954bd87125f718a5c5","author":{"_account_id":25023,"name":"Jonathan Rosser","email":"jonathan.rosser@rd.bbc.co.uk","username":"jrosser"},"date":"2019-07-16 14:22:03.000000000","message":"Patch Set 4: Code-Review-1","accounts_in_message":[],"_revision_number":4},{"id":"35de79f04cf403c890896d30f1ba9d4f93c41895","author":{"_account_id":17130,"name":"melissaml","email":"ma.lei@99cloud.net","username":"malei"},"date":"2019-07-17 03:28:14.000000000","message":"Uploaded patch set 5.","accounts_in_message":[],"_revision_number":5},{"id":"d11bf8f88fcfdec67acd32734db99562366b2b53","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2019-07-17 05:23:00.000000000","message":"Patch Set 5: Verified+1\n\nBuild succeeded (check pipeline).\n\n- openstack-ansible-linters http://logs.openstack.org/43/657243/5/check/openstack-ansible-linters/4f9ed3f/ : SUCCESS in 4m 37s\n- openstack-ansible-deploy-aio_metal-debian-stable http://logs.openstack.org/43/657243/5/check/openstack-ansible-deploy-aio_metal-debian-stable/2ecd3f3/ : SUCCESS in 1h 16m 17s\n- openstack-ansible-deploy-aio_metal-centos-7 http://logs.openstack.org/43/657243/5/check/openstack-ansible-deploy-aio_metal-centos-7/817e831/ : SUCCESS in 1h 44m 49s\n- openstack-ansible-deploy-aio_metal-opensuse-15 http://logs.openstack.org/43/657243/5/check/openstack-ansible-deploy-aio_metal-opensuse-15/db2275d/ : SUCCESS in 1h 26m 26s\n- openstack-ansible-deploy-aio_metal-ubuntu-bionic http://logs.openstack.org/43/657243/5/check/openstack-ansible-deploy-aio_metal-ubuntu-bionic/a05a162/ : SUCCESS in 1h 06m 53s\n- openstack-ansible-deploy-aio_distro_metal-centos-7 http://logs.openstack.org/43/657243/5/check/openstack-ansible-deploy-aio_distro_metal-centos-7/48eb160/ : SUCCESS in 1h 31m 10s\n- openstack-ansible-deploy-aio_distro_metal-opensuse-15 http://logs.openstack.org/43/657243/5/check/openstack-ansible-deploy-aio_distro_metal-opensuse-15/c4b47e1/ : FAILURE in 44m 57s (non-voting)\n- openstack-ansible-deploy-aio_telemetry_metal-centos-7 http://logs.openstack.org/43/657243/5/check/openstack-ansible-deploy-aio_telemetry_metal-centos-7/b5c8c95/ : SUCCESS in 1h 43m 06s\n- openstack-ansible-deploy-aio_telemetry_metal-debian-stable http://logs.openstack.org/43/657243/5/check/openstack-ansible-deploy-aio_telemetry_metal-debian-stable/5d49584/ : SUCCESS in 1h 14m 32s\n- openstack-ansible-deploy-aio_telemetry_metal-ubuntu-bionic http://logs.openstack.org/43/657243/5/check/openstack-ansible-deploy-aio_telemetry_metal-ubuntu-bionic/3e0851b/ : SUCCESS in 1h 09m 54s\n- openstack-tox-docs http://logs.openstack.org/43/657243/5/check/openstack-tox-docs/8586630/html/ : SUCCESS in 2m 17s\n- build-openstack-releasenotes http://logs.openstack.org/43/657243/5/check/build-openstack-releasenotes/319d567/html/ : SUCCESS in 2m 33s","accounts_in_message":[],"_revision_number":5},{"id":"52a8b6cec5b1a72ad11b760a08c437a679c0d877","author":{"_account_id":28619,"name":"Dmitriy Rabotyagov","email":"noonedeadpunk@gmail.com","username":"noonedeadpunk"},"date":"2019-07-17 05:37:57.000000000","message":"Patch Set 5: Code-Review+2","accounts_in_message":[],"_revision_number":5},{"id":"67074ceaaf78ed85c7b669165d0e4677e261bc84","author":{"_account_id":18279,"name":"Jonathan Herlin","email":"jonte@jherlin.se","username":"Jonher937"},"date":"2019-07-17 06:33:43.000000000","message":"Patch Set 5: Code-Review+1","accounts_in_message":[],"_revision_number":5},{"id":"adcf49ba7e5eb935f511dde158a78df723601720","author":{"_account_id":25023,"name":"Jonathan Rosser","email":"jonathan.rosser@rd.bbc.co.uk","username":"jrosser"},"date":"2019-07-17 15:21:12.000000000","message":"Patch Set 5: Code-Review+2 Workflow+1","accounts_in_message":[],"_revision_number":5},{"id":"af175580cb8a6024d480054a0f92f5b5728719e9","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2019-07-17 15:21:24.000000000","message":"Patch Set 5: -Verified\n\nStarting gate jobs.","accounts_in_message":[],"_revision_number":5},{"id":"f75d6e210e545d0aed74408827af0f96f10d6e7e","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2019-07-17 17:26:02.000000000","message":"Patch Set 5: Verified+2\n\nBuild succeeded (gate pipeline).\n\n- openstack-ansible-linters http://logs.openstack.org/43/657243/5/gate/openstack-ansible-linters/5bcbf64/ : SUCCESS in 6m 49s\n- openstack-ansible-deploy-aio_metal-debian-stable http://logs.openstack.org/43/657243/5/gate/openstack-ansible-deploy-aio_metal-debian-stable/02adcdc/ : SUCCESS in 1h 09m 39s\n- openstack-ansible-deploy-aio_metal-centos-7 http://logs.openstack.org/43/657243/5/gate/openstack-ansible-deploy-aio_metal-centos-7/ec78a46/ : SUCCESS in 1h 32m 37s\n- openstack-ansible-deploy-aio_metal-opensuse-15 http://logs.openstack.org/43/657243/5/gate/openstack-ansible-deploy-aio_metal-opensuse-15/db52c88/ : SUCCESS in 1h 31m 27s\n- openstack-ansible-deploy-aio_metal-ubuntu-bionic http://logs.openstack.org/43/657243/5/gate/openstack-ansible-deploy-aio_metal-ubuntu-bionic/2e4a947/ : SUCCESS in 1h 30m 19s\n- openstack-ansible-deploy-aio_distro_metal-centos-7 http://logs.openstack.org/43/657243/5/gate/openstack-ansible-deploy-aio_distro_metal-centos-7/e80eaaa/ : SUCCESS in 1h 13m 30s\n- openstack-ansible-deploy-aio_telemetry_metal-centos-7 http://logs.openstack.org/43/657243/5/gate/openstack-ansible-deploy-aio_telemetry_metal-centos-7/e46e02f/ : SUCCESS in 1h 58m 36s\n- openstack-ansible-deploy-aio_telemetry_metal-debian-stable http://logs.openstack.org/43/657243/5/gate/openstack-ansible-deploy-aio_telemetry_metal-debian-stable/959346a/ : SUCCESS in 1h 16m 19s\n- openstack-ansible-deploy-aio_telemetry_metal-ubuntu-bionic http://logs.openstack.org/43/657243/5/gate/openstack-ansible-deploy-aio_telemetry_metal-ubuntu-bionic/7898324/ : SUCCESS in 1h 19m 59s\n- openstack-tox-docs http://logs.openstack.org/43/657243/5/gate/openstack-tox-docs/5b06240/html/ : SUCCESS in 2m 48s\n- build-openstack-releasenotes http://logs.openstack.org/43/657243/5/gate/build-openstack-releasenotes/bd484a7/html/ : SUCCESS in 4m 34s","accounts_in_message":[],"_revision_number":5},{"id":"1aa79236ad70b7b1e4a2e94ed0afa1127de7e9da","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2019-07-17 17:26:02.000000000","message":"Change has been successfully merged by Zuul","accounts_in_message":[],"_revision_number":5}],"current_revision_number":5,"current_revision":"be0066587849b37316aca747de20f41307534bdf","revisions":{"690f60c7df6d43efea1e585fd598fc182a518572":{"kind":"REWORK","_number":1,"created":"2019-05-05 12:04:18.000000000","uploader":{"_account_id":17130,"name":"melissaml","email":"ma.lei@99cloud.net","username":"malei"},"ref":"refs/changes/43/657243/1","fetch":{"anonymous http":{"url":"https://review.opendev.org/openstack/openstack-ansible-os_aodh","ref":"refs/changes/43/657243/1","commands":{"Checkout":"git fetch https://review.opendev.org/openstack/openstack-ansible-os_aodh refs/changes/43/657243/1 \u0026\u0026 git checkout FETCH_HEAD","Cherry Pick":"git fetch https://review.opendev.org/openstack/openstack-ansible-os_aodh refs/changes/43/657243/1 \u0026\u0026 git cherry-pick FETCH_HEAD","Format Patch":"git fetch https://review.opendev.org/openstack/openstack-ansible-os_aodh refs/changes/43/657243/1 \u0026\u0026 git format-patch -1 --stdout FETCH_HEAD","Pull":"git pull https://review.opendev.org/openstack/openstack-ansible-os_aodh refs/changes/43/657243/1"}}},"commit":{"parents":[{"commit":"5041b72c7b6a123cb3afcdcda65066252affad44","subject":"Merge \"fix tox python3 overrides\"","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/openstack-ansible-os_aodh/commit/5041b72c7b6a123cb3afcdcda65066252affad44"}]}],"author":{"name":"melissaml","email":"ma.lei@99cloud.net","date":"2019-05-05 12:01:53.000000000","tz":480},"committer":{"name":"melissaml","email":"ma.lei@99cloud.net","date":"2019-05-05 12:03:08.000000000","tz":480},"subject":"Replace git.openstack.org URLs with opendev.org URLs","message":"Replace git.openstack.org URLs with opendev.org URLs\n\nChange-Id: I1231b4d0d0b0865eaf88d42ad3a4c69aa1c2a961\n","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/openstack-ansible-os_aodh/commit/690f60c7df6d43efea1e585fd598fc182a518572"}],"resolve_conflicts_web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/openstack-ansible-os_aodh/commit/690f60c7df6d43efea1e585fd598fc182a518572"}]},"branch":"refs/heads/master"},"5e38c944338587d1afb52c2ebd7f27db90fd31ad":{"kind":"REWORK","_number":2,"created":"2019-05-06 10:48:47.000000000","uploader":{"_account_id":17130,"name":"melissaml","email":"ma.lei@99cloud.net","username":"malei"},"ref":"refs/changes/43/657243/2","fetch":{"anonymous http":{"url":"https://review.opendev.org/openstack/openstack-ansible-os_aodh","ref":"refs/changes/43/657243/2","commands":{"Checkout":"git fetch https://review.opendev.org/openstack/openstack-ansible-os_aodh refs/changes/43/657243/2 \u0026\u0026 git checkout FETCH_HEAD","Cherry Pick":"git fetch https://review.opendev.org/openstack/openstack-ansible-os_aodh refs/changes/43/657243/2 \u0026\u0026 git cherry-pick FETCH_HEAD","Format Patch":"git fetch https://review.opendev.org/openstack/openstack-ansible-os_aodh refs/changes/43/657243/2 \u0026\u0026 git format-patch -1 --stdout FETCH_HEAD","Pull":"git pull https://review.opendev.org/openstack/openstack-ansible-os_aodh refs/changes/43/657243/2"}}},"commit":{"parents":[{"commit":"5041b72c7b6a123cb3afcdcda65066252affad44","subject":"Merge \"fix tox python3 overrides\"","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/openstack-ansible-os_aodh/commit/5041b72c7b6a123cb3afcdcda65066252affad44"}]}],"author":{"name":"melissaml","email":"ma.lei@99cloud.net","date":"2019-05-05 12:01:53.000000000","tz":480},"committer":{"name":"melissaml","email":"ma.lei@99cloud.net","date":"2019-05-06 10:48:36.000000000","tz":480},"subject":"Replace git.openstack.org URLs with opendev.org URLs","message":"Replace git.openstack.org URLs with opendev.org URLs\n\nChange-Id: I1231b4d0d0b0865eaf88d42ad3a4c69aa1c2a961\n","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/openstack-ansible-os_aodh/commit/5e38c944338587d1afb52c2ebd7f27db90fd31ad"}],"resolve_conflicts_web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/openstack-ansible-os_aodh/commit/5e38c944338587d1afb52c2ebd7f27db90fd31ad"}]},"branch":"refs/heads/master"},"4ca2f682631db807eb7984be50fdf53ba6a2b444":{"kind":"REWORK","_number":3,"created":"2019-07-10 14:38:03.000000000","uploader":{"_account_id":17130,"name":"melissaml","email":"ma.lei@99cloud.net","username":"malei"},"ref":"refs/changes/43/657243/3","fetch":{"anonymous http":{"url":"https://review.opendev.org/openstack/openstack-ansible-os_aodh","ref":"refs/changes/43/657243/3","commands":{"Checkout":"git fetch https://review.opendev.org/openstack/openstack-ansible-os_aodh refs/changes/43/657243/3 \u0026\u0026 git checkout FETCH_HEAD","Cherry Pick":"git fetch https://review.opendev.org/openstack/openstack-ansible-os_aodh refs/changes/43/657243/3 \u0026\u0026 git cherry-pick FETCH_HEAD","Format Patch":"git fetch https://review.opendev.org/openstack/openstack-ansible-os_aodh refs/changes/43/657243/3 \u0026\u0026 git format-patch -1 --stdout FETCH_HEAD","Pull":"git pull https://review.opendev.org/openstack/openstack-ansible-os_aodh refs/changes/43/657243/3"}}},"commit":{"parents":[{"commit":"5041b72c7b6a123cb3afcdcda65066252affad44","subject":"Merge \"fix tox python3 overrides\"","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/openstack-ansible-os_aodh/commit/5041b72c7b6a123cb3afcdcda65066252affad44"}]}],"author":{"name":"melissaml","email":"ma.lei@99cloud.net","date":"2019-05-05 12:01:53.000000000","tz":480},"committer":{"name":"melissaml","email":"ma.lei@99cloud.net","date":"2019-07-10 14:37:18.000000000","tz":480},"subject":"Replace git.openstack.org URLs with opendev.org URLs","message":"Replace git.openstack.org URLs with opendev.org URLs\n\nChange-Id: I1231b4d0d0b0865eaf88d42ad3a4c69aa1c2a961\n","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/openstack-ansible-os_aodh/commit/4ca2f682631db807eb7984be50fdf53ba6a2b444"}],"resolve_conflicts_web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/openstack-ansible-os_aodh/commit/4ca2f682631db807eb7984be50fdf53ba6a2b444"}]},"branch":"refs/heads/master"},"433126a5f3319fc0b307cb9845c8de357414a716":{"kind":"REWORK","_number":4,"created":"2019-07-13 19:37:29.000000000","uploader":{"_account_id":17130,"name":"melissaml","email":"ma.lei@99cloud.net","username":"malei"},"ref":"refs/changes/43/657243/4","fetch":{"anonymous http":{"url":"https://review.opendev.org/openstack/openstack-ansible-os_aodh","ref":"refs/changes/43/657243/4","commands":{"Checkout":"git fetch https://review.opendev.org/openstack/openstack-ansible-os_aodh refs/changes/43/657243/4 \u0026\u0026 git checkout FETCH_HEAD","Cherry Pick":"git fetch https://review.opendev.org/openstack/openstack-ansible-os_aodh refs/changes/43/657243/4 \u0026\u0026 git cherry-pick FETCH_HEAD","Format Patch":"git fetch https://review.opendev.org/openstack/openstack-ansible-os_aodh refs/changes/43/657243/4 \u0026\u0026 git format-patch -1 --stdout FETCH_HEAD","Pull":"git pull https://review.opendev.org/openstack/openstack-ansible-os_aodh refs/changes/43/657243/4"}}},"commit":{"parents":[{"commit":"5041b72c7b6a123cb3afcdcda65066252affad44","subject":"Merge \"fix tox python3 overrides\"","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/openstack-ansible-os_aodh/commit/5041b72c7b6a123cb3afcdcda65066252affad44"}]}],"author":{"name":"melissaml","email":"ma.lei@99cloud.net","date":"2019-05-05 12:01:53.000000000","tz":480},"committer":{"name":"melissaml","email":"ma.lei@99cloud.net","date":"2019-07-13 19:37:14.000000000","tz":480},"subject":"Replace git.openstack.org URLs with opendev.org URLs","message":"Replace git.openstack.org URLs with opendev.org URLs\n\nChange-Id: I1231b4d0d0b0865eaf88d42ad3a4c69aa1c2a961\n","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/openstack-ansible-os_aodh/commit/433126a5f3319fc0b307cb9845c8de357414a716"}],"resolve_conflicts_web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/openstack-ansible-os_aodh/commit/433126a5f3319fc0b307cb9845c8de357414a716"}]},"branch":"refs/heads/master"},"be0066587849b37316aca747de20f41307534bdf":{"kind":"REWORK","_number":5,"created":"2019-07-17 03:28:14.000000000","uploader":{"_account_id":17130,"name":"melissaml","email":"ma.lei@99cloud.net","username":"malei"},"ref":"refs/changes/43/657243/5","fetch":{"anonymous http":{"url":"https://review.opendev.org/openstack/openstack-ansible-os_aodh","ref":"refs/changes/43/657243/5","commands":{"Checkout":"git fetch https://review.opendev.org/openstack/openstack-ansible-os_aodh refs/changes/43/657243/5 \u0026\u0026 git checkout FETCH_HEAD","Cherry Pick":"git fetch https://review.opendev.org/openstack/openstack-ansible-os_aodh refs/changes/43/657243/5 \u0026\u0026 git cherry-pick FETCH_HEAD","Format Patch":"git fetch https://review.opendev.org/openstack/openstack-ansible-os_aodh refs/changes/43/657243/5 \u0026\u0026 git format-patch -1 --stdout FETCH_HEAD","Pull":"git pull https://review.opendev.org/openstack/openstack-ansible-os_aodh refs/changes/43/657243/5"}}},"commit":{"parents":[{"commit":"5041b72c7b6a123cb3afcdcda65066252affad44","subject":"Merge \"fix tox python3 overrides\"","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/openstack-ansible-os_aodh/commit/5041b72c7b6a123cb3afcdcda65066252affad44"}]}],"author":{"name":"melissaml","email":"ma.lei@99cloud.net","date":"2019-05-05 12:01:53.000000000","tz":480},"committer":{"name":"melissaml","email":"ma.lei@99cloud.net","date":"2019-07-17 03:27:53.000000000","tz":480},"subject":"Replace git.openstack.org URLs with opendev.org URLs","message":"Replace git.openstack.org URLs with opendev.org URLs\n\nChange-Id: I1231b4d0d0b0865eaf88d42ad3a4c69aa1c2a961\n","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/openstack-ansible-os_aodh/commit/be0066587849b37316aca747de20f41307534bdf"}],"resolve_conflicts_web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/openstack-ansible-os_aodh/commit/be0066587849b37316aca747de20f41307534bdf"}]},"branch":"refs/heads/master"}},"requirements":[],"submit_records":[],"submit_requirements":[]}
