)]}'
{"id":"openstack%2Fopenstack-ansible-os_aodh~658012","triplet_id":"openstack%2Fopenstack-ansible-os_aodh~master~Ifdb464469f12fc60ac57ecb6b703c554da836daf","project":"openstack/openstack-ansible-os_aodh","branch":"master","topic":"openstack/openstack-ansible-tests/sync-tests","hashtags":[],"change_id":"Ifdb464469f12fc60ac57ecb6b703c554da836daf","subject":"Updated from OpenStack Ansible Tests","status":"MERGED","created":"2019-05-09 11:34:09.000000000","updated":"2019-05-09 15:07:29.000000000","submitted":"2019-05-09 15:07:29.000000000","submitter":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"total_comment_count":0,"unresolved_comment_count":0,"has_review_started":true,"submission_id":"658012-1557414449301-84f15ffa","meta_rev_id":"aa781d3393871fdd9b7a91fadb84d3ad6ee0df81","_number":658012,"virtual_id_number":658012,"owner":{"_account_id":11131,"name":"OpenStack Proposal Bot","username":"proposal-bot","tags":["SERVICE_USER"]},"actions":{},"labels":{"Verified":{"approved":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"all":[{"value":0,"_account_id":17799,"name":"Logan V","email":"logan2211@gmail.com","username":"Logan2211"},{"value":2,"date":"2019-05-09 15:07:29.000000000","permitted_voting_range":{"min":2,"max":2},"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]}],"values":{"-2":"Fails","-1":"Doesn\u0027t seem to work"," 0":"No score","+1":"Works for me","+2":"Verified"},"description":"","default_value":0,"optional":true},"Code-Review":{"approved":{"_account_id":17799,"name":"Logan V","email":"logan2211@gmail.com","username":"Logan2211"},"all":[{"value":2,"date":"2019-05-09 13:00:01.000000000","permitted_voting_range":{"min":2,"max":2},"_account_id":17799,"name":"Logan V","email":"logan2211@gmail.com","username":"Logan2211"},{"value":0,"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]}],"values":{"-2":"Do not merge","-1":"This patch needs further work before it can be merged"," 0":"No score","+1":"Looks good to me, but someone else must approve","+2":"Looks good to me (core reviewer)"},"description":"","default_value":0,"optional":true},"Workflow":{"approved":{"_account_id":17799,"name":"Logan V","email":"logan2211@gmail.com","username":"Logan2211"},"all":[{"value":1,"date":"2019-05-09 13:00:01.000000000","permitted_voting_range":{"min":1,"max":1},"_account_id":17799,"name":"Logan V","email":"logan2211@gmail.com","username":"Logan2211"},{"value":0,"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]}],"values":{"-1":"Work in progress"," 0":"Ready for reviews","+1":"Approved"},"description":"","default_value":0,"optional":true},"Backport-Candidate":{"all":[{"value":0,"_account_id":17799,"name":"Logan V","email":"logan2211@gmail.com","username":"Logan2211"},{"value":0,"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]}],"values":{"-1":"Not A Backport Candidate"," 0":"Backport Review Needed","+1":"Proposed Backport"},"description":"","default_value":0,"optional":true}},"removable_reviewers":[],"reviewers":{"REVIEWER":[{"_account_id":17799,"name":"Logan V","email":"logan2211@gmail.com","username":"Logan2211"},{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]}]},"pending_reviewers":{},"reviewer_updates":[{"updated":"2019-05-09 13:00:01.000000000","updated_by":{"_account_id":17799,"name":"Logan V","email":"logan2211@gmail.com","username":"Logan2211"},"reviewer":{"_account_id":17799,"name":"Logan V","email":"logan2211@gmail.com","username":"Logan2211"},"state":"REVIEWER"},{"updated":"2019-05-09 15:07:29.000000000","updated_by":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"reviewer":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"state":"REVIEWER"}],"messages":[{"id":"258697f764f84327a3f2fe61ba470de5e159dee9","author":{"_account_id":11131,"name":"OpenStack Proposal Bot","username":"proposal-bot","tags":["SERVICE_USER"]},"date":"2019-05-09 11:34:09.000000000","message":"Uploaded patch set 1.","accounts_in_message":[],"_revision_number":1},{"id":"5a6fdf98e452f574bbd98da21cf217c464bcc0a3","author":{"_account_id":17799,"name":"Logan V","email":"logan2211@gmail.com","username":"Logan2211"},"date":"2019-05-09 13:00:01.000000000","message":"Patch Set 1: Code-Review+2 Workflow+1","accounts_in_message":[],"_revision_number":1},{"id":"958212c6d5a3a83971371240a9ef64534f47217a","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2019-05-09 13:10:41.000000000","message":"Patch Set 1: Verified+1\n\nBuild succeeded (check pipeline).\n\n- openstack-ansible-linters http://logs.openstack.org/12/658012/1/check/openstack-ansible-linters/40b8313/ : SUCCESS in 5m 42s\n- openstack-ansible-deploy-aio_telemetry_metal-centos-7 http://logs.openstack.org/12/658012/1/check/openstack-ansible-deploy-aio_telemetry_metal-centos-7/2ac6c3b/ : SUCCESS in 1h 28m 26s\n- openstack-ansible-deploy-aio_telemetry_metal-debian-stable http://logs.openstack.org/12/658012/1/check/openstack-ansible-deploy-aio_telemetry_metal-debian-stable/afb4d61/ : SUCCESS in 1h 22m 54s\n- openstack-ansible-deploy-aio_telemetry_metal-ubuntu-bionic http://logs.openstack.org/12/658012/1/check/openstack-ansible-deploy-aio_telemetry_metal-ubuntu-bionic/27c953c/ : SUCCESS in 1h 09m 48s\n- openstack-tox-docs http://logs.openstack.org/12/658012/1/check/openstack-tox-docs/d5fb06a/html/ : SUCCESS in 2m 27s\n- build-openstack-releasenotes http://logs.openstack.org/12/658012/1/check/build-openstack-releasenotes/369c76a/html/ : SUCCESS in 2m 10s","accounts_in_message":[],"_revision_number":1},{"id":"6f785f1c6dbdf26ae006bd6aec5d1ce727cfc537","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2019-05-09 13:10:53.000000000","message":"Patch Set 1: -Verified\n\nStarting gate jobs.","accounts_in_message":[],"_revision_number":1},{"id":"81c5037677f11fb0c3fd2ac9acc11d2d5b48bbb7","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2019-05-09 15:07:29.000000000","message":"Change has been successfully merged by Zuul","accounts_in_message":[],"_revision_number":1},{"id":"fe823fdabe7c44525d2cebef3a8dcbe9a1f462db","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2019-05-09 15:07:29.000000000","message":"Patch Set 1: Verified+2\n\nBuild succeeded (gate pipeline).\n\n- openstack-ansible-linters http://logs.openstack.org/12/658012/1/gate/openstack-ansible-linters/820de32/ : SUCCESS in 5m 17s\n- openstack-ansible-deploy-aio_telemetry_metal-centos-7 http://logs.openstack.org/12/658012/1/gate/openstack-ansible-deploy-aio_telemetry_metal-centos-7/b499251/ : SUCCESS in 1h 49m 52s\n- openstack-ansible-deploy-aio_telemetry_metal-debian-stable http://logs.openstack.org/12/658012/1/gate/openstack-ansible-deploy-aio_telemetry_metal-debian-stable/d8fe9c8/ : SUCCESS in 1h 12m 52s\n- openstack-ansible-deploy-aio_telemetry_metal-ubuntu-bionic http://logs.openstack.org/12/658012/1/gate/openstack-ansible-deploy-aio_telemetry_metal-ubuntu-bionic/7483564/ : SUCCESS in 1h 06m 08s\n- openstack-tox-docs http://logs.openstack.org/12/658012/1/gate/openstack-tox-docs/344a23f/html/ : SUCCESS in 2m 40s\n- build-openstack-releasenotes http://logs.openstack.org/12/658012/1/gate/build-openstack-releasenotes/30b137a/html/ : SUCCESS in 2m 51s","accounts_in_message":[],"_revision_number":1}],"current_revision_number":1,"current_revision":"07ea68ebfa170c187ed9320e596ef744d7ac8277","revisions":{"07ea68ebfa170c187ed9320e596ef744d7ac8277":{"kind":"REWORK","_number":1,"created":"2019-05-09 11:34:09.000000000","uploader":{"_account_id":11131,"name":"OpenStack Proposal Bot","username":"proposal-bot","tags":["SERVICE_USER"]},"ref":"refs/changes/12/658012/1","fetch":{"anonymous http":{"url":"https://review.opendev.org/openstack/openstack-ansible-os_aodh","ref":"refs/changes/12/658012/1","commands":{"Checkout":"git fetch https://review.opendev.org/openstack/openstack-ansible-os_aodh refs/changes/12/658012/1 \u0026\u0026 git checkout FETCH_HEAD","Cherry Pick":"git fetch https://review.opendev.org/openstack/openstack-ansible-os_aodh refs/changes/12/658012/1 \u0026\u0026 git cherry-pick FETCH_HEAD","Format Patch":"git fetch https://review.opendev.org/openstack/openstack-ansible-os_aodh refs/changes/12/658012/1 \u0026\u0026 git format-patch -1 --stdout FETCH_HEAD","Pull":"git pull https://review.opendev.org/openstack/openstack-ansible-os_aodh refs/changes/12/658012/1"}}},"commit":{"parents":[{"commit":"5041b72c7b6a123cb3afcdcda65066252affad44","subject":"Merge \"fix tox python3 overrides\"","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/openstack-ansible-os_aodh/commit/5041b72c7b6a123cb3afcdcda65066252affad44"}]}],"author":{"name":"OpenStack Proposal Bot","email":"openstack-infra@lists.openstack.org","date":"2019-05-09 11:34:08.000000000","tz":0},"committer":{"name":"OpenStack Proposal Bot","email":"openstack-infra@lists.openstack.org","date":"2019-05-09 11:34:08.000000000","tz":0},"subject":"Updated from OpenStack Ansible Tests","message":"Updated from OpenStack Ansible Tests\n\nChange-Id: Ifdb464469f12fc60ac57ecb6b703c554da836daf\n","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/openstack-ansible-os_aodh/commit/07ea68ebfa170c187ed9320e596ef744d7ac8277"}],"resolve_conflicts_web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/openstack-ansible-os_aodh/commit/07ea68ebfa170c187ed9320e596ef744d7ac8277"}]},"branch":"refs/heads/master"}},"requirements":[],"submit_records":[],"submit_requirements":[]}
