)]}'
{"id":"openstack%2Fopenstack-ansible-os_aodh~658376","triplet_id":"openstack%2Fopenstack-ansible-os_aodh~master~I71b829b9b6328a1810b4c084859d1e702f64e2c8","project":"openstack/openstack-ansible-os_aodh","branch":"master","topic":"use-integrated-repo-for-all-tests","hashtags":[],"change_id":"I71b829b9b6328a1810b4c084859d1e702f64e2c8","subject":"Add integrated coverage","status":"MERGED","created":"2019-05-10 14:08:13.000000000","updated":"2019-05-15 12:10:10.000000000","submitted":"2019-05-15 12:10:10.000000000","submitter":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"total_comment_count":2,"unresolved_comment_count":0,"has_review_started":true,"submission_id":"658376-1557922210512-b2ea1c39","meta_rev_id":"d5c41bc82b9bd260def92a5000c5e42e36a5f243","_number":658376,"virtual_id_number":658376,"owner":{"_account_id":17068,"name":"Jean-Philippe Evrard","email":"openstack@a.spamming.party","username":"evrardjp"},"actions":{},"labels":{"Verified":{"approved":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"all":[{"value":0,"_account_id":6816,"name":"Jesse Pretorius","email":"jesse@odyssey4.me","username":"jesse-pretorius"},{"value":0,"_account_id":28619,"name":"Dmitriy Rabotyagov","email":"noonedeadpunk@gmail.com","username":"noonedeadpunk"},{"value":2,"date":"2019-05-15 12:10:10.000000000","permitted_voting_range":{"min":2,"max":2},"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},{"value":0,"_account_id":17068,"name":"Jean-Philippe Evrard","email":"openstack@a.spamming.party","username":"evrardjp"},{"value":0,"_account_id":25023,"name":"Jonathan Rosser","email":"jonathan.rosser@rd.bbc.co.uk","username":"jrosser"}],"values":{"-2":"Fails","-1":"Doesn\u0027t seem to work"," 0":"No score","+1":"Works for me","+2":"Verified"},"description":"","default_value":0,"optional":true},"Code-Review":{"approved":{"_account_id":28619,"name":"Dmitriy Rabotyagov","email":"noonedeadpunk@gmail.com","username":"noonedeadpunk"},"all":[{"value":0,"_account_id":6816,"name":"Jesse Pretorius","email":"jesse@odyssey4.me","username":"jesse-pretorius"},{"value":2,"date":"2019-05-15 09:49:57.000000000","permitted_voting_range":{"min":2,"max":2},"_account_id":28619,"name":"Dmitriy Rabotyagov","email":"noonedeadpunk@gmail.com","username":"noonedeadpunk"},{"value":0,"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},{"value":0,"_account_id":17068,"name":"Jean-Philippe Evrard","email":"openstack@a.spamming.party","username":"evrardjp"},{"value":2,"date":"2019-05-14 21:21:08.000000000","permitted_voting_range":{"min":2,"max":2},"_account_id":25023,"name":"Jonathan Rosser","email":"jonathan.rosser@rd.bbc.co.uk","username":"jrosser"}],"values":{"-2":"Do not merge","-1":"This patch needs further work before it can be merged"," 0":"No score","+1":"Looks good to me, but someone else must approve","+2":"Looks good to me (core reviewer)"},"description":"","default_value":0,"optional":true},"Workflow":{"approved":{"_account_id":28619,"name":"Dmitriy Rabotyagov","email":"noonedeadpunk@gmail.com","username":"noonedeadpunk"},"all":[{"value":0,"_account_id":6816,"name":"Jesse Pretorius","email":"jesse@odyssey4.me","username":"jesse-pretorius"},{"value":1,"date":"2019-05-15 09:49:57.000000000","permitted_voting_range":{"min":1,"max":1},"_account_id":28619,"name":"Dmitriy Rabotyagov","email":"noonedeadpunk@gmail.com","username":"noonedeadpunk"},{"value":0,"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},{"value":0,"_account_id":17068,"name":"Jean-Philippe Evrard","email":"openstack@a.spamming.party","username":"evrardjp"},{"value":0,"_account_id":25023,"name":"Jonathan Rosser","email":"jonathan.rosser@rd.bbc.co.uk","username":"jrosser"}],"values":{"-1":"Work in progress"," 0":"Ready for reviews","+1":"Approved"},"description":"","default_value":0,"optional":true},"Backport-Candidate":{"all":[{"value":0,"_account_id":6816,"name":"Jesse Pretorius","email":"jesse@odyssey4.me","username":"jesse-pretorius"},{"value":0,"_account_id":28619,"name":"Dmitriy Rabotyagov","email":"noonedeadpunk@gmail.com","username":"noonedeadpunk"},{"value":0,"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},{"value":0,"_account_id":17068,"name":"Jean-Philippe Evrard","email":"openstack@a.spamming.party","username":"evrardjp"},{"value":0,"_account_id":25023,"name":"Jonathan Rosser","email":"jonathan.rosser@rd.bbc.co.uk","username":"jrosser"}],"values":{"-1":"Not A Backport Candidate"," 0":"Backport Review Needed","+1":"Proposed Backport"},"description":"","default_value":0,"optional":true}},"removable_reviewers":[],"reviewers":{"REVIEWER":[{"_account_id":6816,"name":"Jesse Pretorius","email":"jesse@odyssey4.me","username":"jesse-pretorius"},{"_account_id":17068,"name":"Jean-Philippe Evrard","email":"openstack@a.spamming.party","username":"evrardjp"},{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},{"_account_id":25023,"name":"Jonathan Rosser","email":"jonathan.rosser@rd.bbc.co.uk","username":"jrosser"},{"_account_id":28619,"name":"Dmitriy Rabotyagov","email":"noonedeadpunk@gmail.com","username":"noonedeadpunk"}]},"pending_reviewers":{},"reviewer_updates":[{"updated":"2019-05-10 14:08:57.000000000","updated_by":{"_account_id":6816,"name":"Jesse Pretorius","email":"jesse@odyssey4.me","username":"jesse-pretorius"},"reviewer":{"_account_id":6816,"name":"Jesse Pretorius","email":"jesse@odyssey4.me","username":"jesse-pretorius"},"state":"REVIEWER"},{"updated":"2019-05-14 21:21:08.000000000","updated_by":{"_account_id":25023,"name":"Jonathan Rosser","email":"jonathan.rosser@rd.bbc.co.uk","username":"jrosser"},"reviewer":{"_account_id":25023,"name":"Jonathan Rosser","email":"jonathan.rosser@rd.bbc.co.uk","username":"jrosser"},"state":"REVIEWER"},{"updated":"2019-05-15 09:49:57.000000000","updated_by":{"_account_id":28619,"name":"Dmitriy Rabotyagov","email":"noonedeadpunk@gmail.com","username":"noonedeadpunk"},"reviewer":{"_account_id":28619,"name":"Dmitriy Rabotyagov","email":"noonedeadpunk@gmail.com","username":"noonedeadpunk"},"state":"REVIEWER"},{"updated":"2019-05-15 12:10:10.000000000","updated_by":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"reviewer":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"state":"REVIEWER"}],"messages":[{"id":"1d1a27df63ed80fa18cea24896b724d6f7ad3381","author":{"_account_id":17068,"name":"Jean-Philippe Evrard","email":"openstack@a.spamming.party","username":"evrardjp"},"date":"2019-05-10 14:08:13.000000000","message":"Uploaded patch set 1.","accounts_in_message":[],"_revision_number":1},{"id":"58deecd9e16a6eeffb274b25f92185d869f52054","author":{"_account_id":6816,"name":"Jesse Pretorius","email":"jesse@odyssey4.me","username":"jesse-pretorius"},"date":"2019-05-10 14:08:57.000000000","message":"Patch Set 1: Code-Review-1\n\n(1 comment)","accounts_in_message":[],"_revision_number":1},{"id":"46217337e8180ba7dd87fb1ceb1d14e90245af24","author":{"_account_id":17068,"name":"Jean-Philippe Evrard","email":"openstack@a.spamming.party","username":"evrardjp"},"date":"2019-05-10 14:48:10.000000000","message":"Patch Set 1:\n\n(1 comment)","accounts_in_message":[],"_revision_number":1},{"id":"c549d9e4baa31a7192a184e60fa66824ff8d3866","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2019-05-10 15:59:36.000000000","message":"Patch Set 1: Verified+1\n\nBuild succeeded (check pipeline).\n\n- openstack-ansible-linters http://logs.openstack.org/76/658376/1/check/openstack-ansible-linters/d9db367/ : SUCCESS in 5m 36s\n- openstack-ansible-deploy-aio_metal-debian-stable http://logs.openstack.org/76/658376/1/check/openstack-ansible-deploy-aio_metal-debian-stable/bb172e8/ : SUCCESS in 1h 02m 12s\n- openstack-ansible-deploy-aio_metal-centos-7 http://logs.openstack.org/76/658376/1/check/openstack-ansible-deploy-aio_metal-centos-7/3cf4e4b/ : SUCCESS in 1h 16m 38s\n- openstack-ansible-deploy-aio_metal-opensuse-150 http://logs.openstack.org/76/658376/1/check/openstack-ansible-deploy-aio_metal-opensuse-150/5b413c5/ : SUCCESS in 1h 08m 57s\n- openstack-ansible-deploy-aio_metal-ubuntu-bionic http://logs.openstack.org/76/658376/1/check/openstack-ansible-deploy-aio_metal-ubuntu-bionic/6123ca8/ : SUCCESS in 55m 48s\n- openstack-ansible-deploy-aio_distro_metal-centos-7 http://logs.openstack.org/76/658376/1/check/openstack-ansible-deploy-aio_distro_metal-centos-7/9923742/ : FAILURE in 38m 37s (non-voting)\n- openstack-ansible-deploy-aio_telemetry_metal-centos-7 http://logs.openstack.org/76/658376/1/check/openstack-ansible-deploy-aio_telemetry_metal-centos-7/875245b/ : SUCCESS in 1h 43m 49s\n- openstack-ansible-deploy-aio_telemetry_metal-debian-stable http://logs.openstack.org/76/658376/1/check/openstack-ansible-deploy-aio_telemetry_metal-debian-stable/a2d62c2/ : SUCCESS in 1h 17m 09s\n- openstack-ansible-deploy-aio_telemetry_metal-ubuntu-bionic http://logs.openstack.org/76/658376/1/check/openstack-ansible-deploy-aio_telemetry_metal-ubuntu-bionic/acc0bc7/ : SUCCESS in 1h 15m 25s\n- openstack-tox-docs http://logs.openstack.org/76/658376/1/check/openstack-tox-docs/6f071d5/html/ : SUCCESS in 2m 30s","accounts_in_message":[],"_revision_number":1},{"id":"9f59fb31b7894fd6eb1e7b253b23954340397342","author":{"_account_id":17068,"name":"Jean-Philippe Evrard","email":"openstack@a.spamming.party","username":"evrardjp"},"date":"2019-05-11 20:20:27.000000000","message":"Patch Set 1:\n\n\u003e (1 comment)\n\nWith the dynamic scenarios, it should just fine, can we get this through?","accounts_in_message":[],"_revision_number":1},{"id":"438147277c2b82e33373dd1d3d6336ca641b87c9","author":{"_account_id":17068,"name":"Jean-Philippe Evrard","email":"openstack@a.spamming.party","username":"evrardjp"},"date":"2019-05-11 20:21:21.000000000","message":"Patch Set 1:\n\n\u003e \u003e (1 comment)\n \u003e \n \u003e With the dynamic scenarios, it should just fine, can we get this\n \u003e through?\n\nI would be fine if we split this in two, one to move to the dynamic jobs, one to validate suse is working on those. But I prefer we have a common approach.","accounts_in_message":[],"_revision_number":1},{"id":"b4a8ab8bcc41315c2e2fc73dd9f053d3fa84bb8e","author":{"_account_id":17068,"name":"Jean-Philippe Evrard","email":"openstack@a.spamming.party","username":"evrardjp"},"date":"2019-05-14 09:58:00.000000000","message":"Uploaded patch set 2.","accounts_in_message":[],"_revision_number":2},{"id":"4c46ce509fe967f527185fa20e72ee6a6d3b2449","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2019-05-14 10:57:03.000000000","message":"Patch Set 2: Verified-1\n\nBuild failed (check pipeline).  For information on how to proceed, see\nhttp://docs.openstack.org/infra/manual/developers.html#automated-testing\n\n\n- openstack-ansible-linters http://logs.openstack.org/76/658376/2/check/openstack-ansible-linters/ba92372/ : SUCCESS in 4m 04s\n- openstack-ansible-deploy-aio_metal-debian-stable http://logs.openstack.org/76/658376/2/check/openstack-ansible-deploy-aio_metal-debian-stable/b82e70d/ : RETRY_LIMIT in 2m 10s\n- openstack-ansible-deploy-aio_metal-centos-7 http://logs.openstack.org/76/658376/2/check/openstack-ansible-deploy-aio_metal-centos-7/4f96ea9/ : RETRY_LIMIT in 2m 41s\n- openstack-ansible-deploy-aio_metal-opensuse-150 http://logs.openstack.org/76/658376/2/check/openstack-ansible-deploy-aio_metal-opensuse-150/0a3c966/ : RETRY_LIMIT in 2m 22s\n- openstack-ansible-deploy-aio_metal-ubuntu-bionic http://logs.openstack.org/76/658376/2/check/openstack-ansible-deploy-aio_metal-ubuntu-bionic/f261ae1/ : RETRY_LIMIT in 1m 24s\n- openstack-ansible-deploy-aio_distro_metal-centos-7 http://logs.openstack.org/76/658376/2/check/openstack-ansible-deploy-aio_distro_metal-centos-7/5128f3c/ : RETRY_LIMIT in 2m 04s (non-voting)\n- openstack-ansible-deploy-aio_telemetry_metal-centos-7 http://logs.openstack.org/76/658376/2/check/openstack-ansible-deploy-aio_telemetry_metal-centos-7/210110e/ : RETRY_LIMIT in 3m 07s\n- openstack-ansible-deploy-aio_telemetry_metal-debian-stable http://logs.openstack.org/76/658376/2/check/openstack-ansible-deploy-aio_telemetry_metal-debian-stable/10df82e/ : RETRY_LIMIT in 1m 54s\n- openstack-ansible-deploy-aio_telemetry_metal-ubuntu-bionic http://logs.openstack.org/76/658376/2/check/openstack-ansible-deploy-aio_telemetry_metal-ubuntu-bionic/74398eb/ : RETRY_LIMIT in 1m 50s\n- openstack-tox-docs http://logs.openstack.org/76/658376/2/check/openstack-tox-docs/79c0890/html/ : SUCCESS in 2m 51s","accounts_in_message":[],"_revision_number":2},{"id":"87edbe43eab22367da1090852cf2e0d973b55810","author":{"_account_id":28619,"name":"Dmitriy Rabotyagov","email":"noonedeadpunk@gmail.com","username":"noonedeadpunk"},"date":"2019-05-14 13:39:26.000000000","message":"Patch Set 3: Commit message was updated.","accounts_in_message":[],"_revision_number":3},{"id":"acc79a7f9b8232b9123f65efb8e356a9fcc5cc56","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2019-05-14 15:19:11.000000000","message":"Patch Set 3: Verified+1\n\nBuild succeeded (check pipeline).\n\n- openstack-ansible-linters http://logs.openstack.org/76/658376/3/check/openstack-ansible-linters/9ed5230/ : SUCCESS in 4m 58s\n- openstack-ansible-deploy-aio_metal-debian-stable http://logs.openstack.org/76/658376/3/check/openstack-ansible-deploy-aio_metal-debian-stable/f0e4eeb/ : SUCCESS in 1h 04m 39s\n- openstack-ansible-deploy-aio_metal-centos-7 http://logs.openstack.org/76/658376/3/check/openstack-ansible-deploy-aio_metal-centos-7/8539965/ : SUCCESS in 1h 29m 07s\n- openstack-ansible-deploy-aio_metal-opensuse-150 http://logs.openstack.org/76/658376/3/check/openstack-ansible-deploy-aio_metal-opensuse-150/5712407/ : SUCCESS in 1h 08m 18s\n- openstack-ansible-deploy-aio_metal-ubuntu-bionic http://logs.openstack.org/76/658376/3/check/openstack-ansible-deploy-aio_metal-ubuntu-bionic/993f638/ : SUCCESS in 1h 04m 47s\n- openstack-ansible-deploy-aio_distro_metal-centos-7 http://logs.openstack.org/76/658376/3/check/openstack-ansible-deploy-aio_distro_metal-centos-7/95db64a/ : FAILURE in 41m 17s (non-voting)\n- openstack-ansible-deploy-aio_telemetry_metal-centos-7 http://logs.openstack.org/76/658376/3/check/openstack-ansible-deploy-aio_telemetry_metal-centos-7/85c38f7/ : SUCCESS in 1h 34m 03s\n- openstack-ansible-deploy-aio_telemetry_metal-debian-stable http://logs.openstack.org/76/658376/3/check/openstack-ansible-deploy-aio_telemetry_metal-debian-stable/1b2bfdb/ : SUCCESS in 1h 24m 24s\n- openstack-ansible-deploy-aio_telemetry_metal-ubuntu-bionic http://logs.openstack.org/76/658376/3/check/openstack-ansible-deploy-aio_telemetry_metal-ubuntu-bionic/e7b2ba6/ : SUCCESS in 1h 14m 57s\n- openstack-tox-docs http://logs.openstack.org/76/658376/3/check/openstack-tox-docs/f85afb8/html/ : SUCCESS in 2m 48s","accounts_in_message":[],"_revision_number":3},{"id":"893d9c54e2f57b6dd0660bb2e8f8befdbec2f063","author":{"_account_id":28619,"name":"Dmitriy Rabotyagov","email":"noonedeadpunk@gmail.com","username":"noonedeadpunk"},"date":"2019-05-14 21:16:48.000000000","message":"Patch Set 3:\n\nrecheck","accounts_in_message":[],"_revision_number":3},{"id":"5f3725d1a1a35fc734ce75a9581fe3f75d8dbb65","author":{"_account_id":25023,"name":"Jonathan Rosser","email":"jonathan.rosser@rd.bbc.co.uk","username":"jrosser"},"date":"2019-05-14 21:21:08.000000000","message":"Patch Set 3: Code-Review+2","accounts_in_message":[],"_revision_number":3},{"id":"faa17d7eb673bc21c2db07432e8c7b78c0b462be","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2019-05-14 23:16:24.000000000","message":"Patch Set 3:\n\nBuild succeeded (check pipeline).\n\n- openstack-ansible-linters http://logs.openstack.org/76/658376/3/check/openstack-ansible-linters/3e71eb3/ : SUCCESS in 5m 46s\n- openstack-ansible-deploy-aio_metal-debian-stable http://logs.openstack.org/76/658376/3/check/openstack-ansible-deploy-aio_metal-debian-stable/6087059/ : SUCCESS in 1h 10m 51s\n- openstack-ansible-deploy-aio_metal-centos-7 http://logs.openstack.org/76/658376/3/check/openstack-ansible-deploy-aio_metal-centos-7/5ac08de/ : SUCCESS in 1h 34m 05s\n- openstack-ansible-deploy-aio_metal-opensuse-150 http://logs.openstack.org/76/658376/3/check/openstack-ansible-deploy-aio_metal-opensuse-150/78559e9/ : SUCCESS in 1h 18m 05s\n- openstack-ansible-deploy-aio_metal-ubuntu-bionic http://logs.openstack.org/76/658376/3/check/openstack-ansible-deploy-aio_metal-ubuntu-bionic/6fd6b62/ : SUCCESS in 1h 58m 53s\n- openstack-ansible-deploy-aio_distro_metal-centos-7 http://logs.openstack.org/76/658376/3/check/openstack-ansible-deploy-aio_distro_metal-centos-7/aa06116/ : FAILURE in 39m 15s (non-voting)\n- openstack-ansible-deploy-aio_telemetry_metal-centos-7 http://logs.openstack.org/76/658376/3/check/openstack-ansible-deploy-aio_telemetry_metal-centos-7/1f5d90d/ : SUCCESS in 1h 41m 08s\n- openstack-ansible-deploy-aio_telemetry_metal-debian-stable http://logs.openstack.org/76/658376/3/check/openstack-ansible-deploy-aio_telemetry_metal-debian-stable/5cf83e0/ : SUCCESS in 1h 16m 52s\n- openstack-ansible-deploy-aio_telemetry_metal-ubuntu-bionic http://logs.openstack.org/76/658376/3/check/openstack-ansible-deploy-aio_telemetry_metal-ubuntu-bionic/78a60af/ : SUCCESS in 1h 08m 53s\n- openstack-tox-docs http://logs.openstack.org/76/658376/3/check/openstack-tox-docs/96f03ad/html/ : SUCCESS in 3m 47s","accounts_in_message":[],"_revision_number":3},{"id":"a89f4cc9edeefb26fdc9f852ccadae7303816786","author":{"_account_id":28619,"name":"Dmitriy Rabotyagov","email":"noonedeadpunk@gmail.com","username":"noonedeadpunk"},"date":"2019-05-15 09:49:57.000000000","message":"Patch Set 3: Code-Review+2 Workflow+1\n\nHurray, dynamic execution really works!\nWe can drop _telemetry_ jobs now","accounts_in_message":[],"_revision_number":3},{"id":"0e47fd569d52bd7fb806fea78d00019335029cc5","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2019-05-15 09:50:11.000000000","message":"Patch Set 3: -Verified\n\nStarting gate jobs.","accounts_in_message":[],"_revision_number":3},{"id":"1696a6bb12e2b1a9b7eb732159979f4c66b0bc64","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2019-05-15 12:10:10.000000000","message":"Patch Set 3: Verified+2\n\nBuild succeeded (gate pipeline).\n\n- openstack-ansible-linters http://logs.openstack.org/76/658376/3/gate/openstack-ansible-linters/8c9607f/ : SUCCESS in 4m 49s\n- openstack-ansible-deploy-aio_metal-debian-stable http://logs.openstack.org/76/658376/3/gate/openstack-ansible-deploy-aio_metal-debian-stable/3fa196f/ : SUCCESS in 1h 06m 01s\n- openstack-ansible-deploy-aio_metal-centos-7 http://logs.openstack.org/76/658376/3/gate/openstack-ansible-deploy-aio_metal-centos-7/a882a87/ : SUCCESS in 2h 06m 47s\n- openstack-ansible-deploy-aio_metal-opensuse-150 http://logs.openstack.org/76/658376/3/gate/openstack-ansible-deploy-aio_metal-opensuse-150/3711792/ : SUCCESS in 1h 35m 49s\n- openstack-ansible-deploy-aio_metal-ubuntu-bionic http://logs.openstack.org/76/658376/3/gate/openstack-ansible-deploy-aio_metal-ubuntu-bionic/378d1f9/ : SUCCESS in 1h 07m 23s\n- openstack-ansible-deploy-aio_telemetry_metal-centos-7 http://logs.openstack.org/76/658376/3/gate/openstack-ansible-deploy-aio_telemetry_metal-centos-7/2af7670/ : SUCCESS in 1h 41m 16s\n- openstack-ansible-deploy-aio_telemetry_metal-debian-stable http://logs.openstack.org/76/658376/3/gate/openstack-ansible-deploy-aio_telemetry_metal-debian-stable/ed94e05/ : SUCCESS in 1h 15m 07s\n- openstack-ansible-deploy-aio_telemetry_metal-ubuntu-bionic http://logs.openstack.org/76/658376/3/gate/openstack-ansible-deploy-aio_telemetry_metal-ubuntu-bionic/a2a0738/ : SUCCESS in 1h 08m 17s\n- openstack-tox-docs http://logs.openstack.org/76/658376/3/gate/openstack-tox-docs/42f492c/html/ : SUCCESS in 2m 35s","accounts_in_message":[],"_revision_number":3},{"id":"63ba8a2ce0112a6d28cdb5880b938cb7caa07975","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2019-05-15 12:10:10.000000000","message":"Change has been successfully merged by Zuul","accounts_in_message":[],"_revision_number":3}],"current_revision_number":3,"current_revision":"8585f43eee839b115cafa6e15e6fc765a001db89","revisions":{"aacbf98fc7e1207c01319e66e09dd56c11ec4ea2":{"kind":"REWORK","_number":1,"created":"2019-05-10 14:08:13.000000000","uploader":{"_account_id":17068,"name":"Jean-Philippe Evrard","email":"openstack@a.spamming.party","username":"evrardjp"},"ref":"refs/changes/76/658376/1","fetch":{"anonymous http":{"url":"https://review.opendev.org/openstack/openstack-ansible-os_aodh","ref":"refs/changes/76/658376/1","commands":{"Checkout":"git fetch https://review.opendev.org/openstack/openstack-ansible-os_aodh refs/changes/76/658376/1 \u0026\u0026 git checkout FETCH_HEAD","Cherry Pick":"git fetch https://review.opendev.org/openstack/openstack-ansible-os_aodh refs/changes/76/658376/1 \u0026\u0026 git cherry-pick FETCH_HEAD","Format Patch":"git fetch https://review.opendev.org/openstack/openstack-ansible-os_aodh refs/changes/76/658376/1 \u0026\u0026 git format-patch -1 --stdout FETCH_HEAD","Pull":"git pull https://review.opendev.org/openstack/openstack-ansible-os_aodh refs/changes/76/658376/1"}}},"commit":{"parents":[{"commit":"5041b72c7b6a123cb3afcdcda65066252affad44","subject":"Merge \"fix tox python3 overrides\"","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/openstack-ansible-os_aodh/commit/5041b72c7b6a123cb3afcdcda65066252affad44"}]}],"author":{"name":"Jean-Philippe Evrard","email":"jean-philippe@evrard.me","date":"2019-05-10 14:05:12.000000000","tz":120},"committer":{"name":"Jean-Philippe Evrard","email":"jean-philippe@evrard.me","date":"2019-05-10 14:07:56.000000000","tz":120},"subject":"Add integrated coverage","message":"Add integrated coverage\n\nCurrently this role doesn\u0027t test the integrated jobs.\nWhile this role isn\u0027t immediately exercised (the playbook\nis skipped), it is good enough to add coverage of the templates,\nshould we adapt the template in a global fashion.\n\nAdditionally, it is a problem to have different jobs\nfor SUSE support, as we want to use the integrated jobs template\nas a mean to get consistent across roles, so that\nit\u0027s easier to monitor jobs.\n\nChange-Id: I71b829b9b6328a1810b4c084859d1e702f64e2c8\n","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/openstack-ansible-os_aodh/commit/aacbf98fc7e1207c01319e66e09dd56c11ec4ea2"}],"resolve_conflicts_web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/openstack-ansible-os_aodh/commit/aacbf98fc7e1207c01319e66e09dd56c11ec4ea2"}]},"branch":"refs/heads/master"},"fe17d08fda14aa5585f95c4ab8f3cc9018b99ebb":{"kind":"REWORK","_number":2,"created":"2019-05-14 09:58:00.000000000","uploader":{"_account_id":17068,"name":"Jean-Philippe Evrard","email":"openstack@a.spamming.party","username":"evrardjp"},"ref":"refs/changes/76/658376/2","fetch":{"anonymous http":{"url":"https://review.opendev.org/openstack/openstack-ansible-os_aodh","ref":"refs/changes/76/658376/2","commands":{"Checkout":"git fetch https://review.opendev.org/openstack/openstack-ansible-os_aodh refs/changes/76/658376/2 \u0026\u0026 git checkout FETCH_HEAD","Cherry Pick":"git fetch https://review.opendev.org/openstack/openstack-ansible-os_aodh refs/changes/76/658376/2 \u0026\u0026 git cherry-pick FETCH_HEAD","Format Patch":"git fetch https://review.opendev.org/openstack/openstack-ansible-os_aodh refs/changes/76/658376/2 \u0026\u0026 git format-patch -1 --stdout FETCH_HEAD","Pull":"git pull https://review.opendev.org/openstack/openstack-ansible-os_aodh refs/changes/76/658376/2"}}},"commit":{"parents":[{"commit":"5041b72c7b6a123cb3afcdcda65066252affad44","subject":"Merge \"fix tox python3 overrides\"","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/openstack-ansible-os_aodh/commit/5041b72c7b6a123cb3afcdcda65066252affad44"}]}],"author":{"name":"Jean-Philippe Evrard","email":"jean-philippe@evrard.me","date":"2019-05-10 14:05:12.000000000","tz":120},"committer":{"name":"Jean-Philippe Evrard","email":"jean-philippe@evrard.me","date":"2019-05-14 09:56:13.000000000","tz":120},"subject":"Add integrated coverage","message":"Add integrated coverage\n\nCurrently this role doesn\u0027t test the integrated jobs.\nWith the dynamic execution of the roles, this role\nalso tests the telemetry stack in the metal jobs template.\n\nChange-Id: I71b829b9b6328a1810b4c084859d1e702f64e2c8\n","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/openstack-ansible-os_aodh/commit/fe17d08fda14aa5585f95c4ab8f3cc9018b99ebb"}],"resolve_conflicts_web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/openstack-ansible-os_aodh/commit/fe17d08fda14aa5585f95c4ab8f3cc9018b99ebb"}]},"branch":"refs/heads/master"},"8585f43eee839b115cafa6e15e6fc765a001db89":{"kind":"NO_CODE_CHANGE","_number":3,"created":"2019-05-14 13:39:26.000000000","uploader":{"_account_id":28619,"name":"Dmitriy Rabotyagov","email":"noonedeadpunk@gmail.com","username":"noonedeadpunk"},"ref":"refs/changes/76/658376/3","fetch":{"anonymous http":{"url":"https://review.opendev.org/openstack/openstack-ansible-os_aodh","ref":"refs/changes/76/658376/3","commands":{"Checkout":"git fetch https://review.opendev.org/openstack/openstack-ansible-os_aodh refs/changes/76/658376/3 \u0026\u0026 git checkout FETCH_HEAD","Cherry Pick":"git fetch https://review.opendev.org/openstack/openstack-ansible-os_aodh refs/changes/76/658376/3 \u0026\u0026 git cherry-pick FETCH_HEAD","Format Patch":"git fetch https://review.opendev.org/openstack/openstack-ansible-os_aodh refs/changes/76/658376/3 \u0026\u0026 git format-patch -1 --stdout FETCH_HEAD","Pull":"git pull https://review.opendev.org/openstack/openstack-ansible-os_aodh refs/changes/76/658376/3"}}},"commit":{"parents":[{"commit":"5041b72c7b6a123cb3afcdcda65066252affad44","subject":"Merge \"fix tox python3 overrides\"","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/openstack-ansible-os_aodh/commit/5041b72c7b6a123cb3afcdcda65066252affad44"}]}],"author":{"name":"Jean-Philippe Evrard","email":"jean-philippe@evrard.me","date":"2019-05-10 14:05:12.000000000","tz":120},"committer":{"name":"Dmitriy Rabotyagov (noonedeadpunk)","email":"noonedeadpunk@ya.ru","date":"2019-05-14 13:39:10.000000000","tz":0},"subject":"Add integrated coverage","message":"Add integrated coverage\n\nCurrently this role doesn\u0027t test the integrated jobs.\nWith the dynamic execution of the roles, this role\nalso tests the telemetry stack in the metal jobs template.\n\nDepends-On: https://review.opendev.org/659064\nChange-Id: I71b829b9b6328a1810b4c084859d1e702f64e2c8\n","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/openstack-ansible-os_aodh/commit/8585f43eee839b115cafa6e15e6fc765a001db89"}],"resolve_conflicts_web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/openstack-ansible-os_aodh/commit/8585f43eee839b115cafa6e15e6fc765a001db89"}]},"branch":"refs/heads/master"}},"requirements":[],"submit_records":[],"submit_requirements":[]}
