)]}'
{"id":"openstack%2Fopenstack-ansible-os_aodh~658405","triplet_id":"openstack%2Fopenstack-ansible-os_aodh~master~If693c0b5195d21414dc2217c0c04cb3bfe11049e","project":"openstack/openstack-ansible-os_aodh","branch":"master","topic":"dynamic-aio-scenario","hashtags":[],"change_id":"If693c0b5195d21414dc2217c0c04cb3bfe11049e","subject":"Use the dynamic aio scenario","status":"ABANDONED","created":"2019-05-10 15:59:30.000000000","updated":"2019-08-07 15:38:07.000000000","total_comment_count":0,"unresolved_comment_count":0,"has_review_started":true,"meta_rev_id":"54499d44f216248bee10dc57a1c48fa6d7d40806","_number":658405,"virtual_id_number":658405,"owner":{"_account_id":6816,"name":"Jesse Pretorius","email":"jesse@odyssey4.me","username":"jesse-pretorius"},"actions":{},"labels":{"Verified":{"disliked":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"all":[{"date":"2019-05-10 19:26:40.000000000","_account_id":6816,"name":"Jesse Pretorius","email":"jesse@odyssey4.me","username":"jesse-pretorius"},{"_account_id":17068,"name":"Jean-Philippe Evrard","email":"openstack@a.spamming.party","username":"evrardjp"},{"value":-1,"date":"2019-05-10 18:02:16.000000000","permitted_voting_range":{"min":-2,"max":2},"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]}],"values":{"-2":"Fails","-1":"Doesn\u0027t seem to work"," 0":"No score","+1":"Works for me","+2":"Verified"},"description":"","value":-1,"default_value":0,"optional":true},"Code-Review":{"disliked":{"_account_id":17068,"name":"Jean-Philippe Evrard","email":"openstack@a.spamming.party","username":"evrardjp"},"all":[{"value":0,"permitted_voting_range":{"min":-1,"max":1},"_account_id":6816,"name":"Jesse Pretorius","email":"jesse@odyssey4.me","username":"jesse-pretorius"},{"value":-1,"date":"2019-05-14 09:53:42.000000000","permitted_voting_range":{"min":-1,"max":1},"_account_id":17068,"name":"Jean-Philippe Evrard","email":"openstack@a.spamming.party","username":"evrardjp"},{"value":0,"permitted_voting_range":{"min":-1,"max":1},"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]}],"values":{"-2":"Do not merge","-1":"This patch needs further work before it can be merged"," 0":"No score","+1":"Looks good to me, but someone else must approve","+2":"Looks good to me (core reviewer)"},"description":"","value":-1,"default_value":0,"optional":true},"Workflow":{"all":[{"value":0,"permitted_voting_range":{"min":-1,"max":0},"_account_id":6816,"name":"Jesse Pretorius","email":"jesse@odyssey4.me","username":"jesse-pretorius"},{"_account_id":17068,"name":"Jean-Philippe Evrard","email":"openstack@a.spamming.party","username":"evrardjp"},{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]}],"values":{"-1":"Work in progress"," 0":"Ready for reviews","+1":"Approved"},"description":"","default_value":0,"optional":true},"Backport-Candidate":{"all":[{"_account_id":6816,"name":"Jesse Pretorius","email":"jesse@odyssey4.me","username":"jesse-pretorius"},{"_account_id":17068,"name":"Jean-Philippe Evrard","email":"openstack@a.spamming.party","username":"evrardjp"},{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]}],"values":{"-1":"Not A Backport Candidate"," 0":"Backport Review Needed","+1":"Proposed Backport"},"description":"","default_value":0,"optional":true}},"removable_reviewers":[],"reviewers":{"REVIEWER":[{"_account_id":6816,"name":"Jesse Pretorius","email":"jesse@odyssey4.me","username":"jesse-pretorius"},{"_account_id":17068,"name":"Jean-Philippe Evrard","email":"openstack@a.spamming.party","username":"evrardjp"},{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]}]},"pending_reviewers":{},"reviewer_updates":[{"updated":"2019-05-10 18:02:16.000000000","updated_by":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"reviewer":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"state":"REVIEWER"},{"updated":"2019-05-14 09:53:42.000000000","updated_by":{"_account_id":17068,"name":"Jean-Philippe Evrard","email":"openstack@a.spamming.party","username":"evrardjp"},"reviewer":{"_account_id":17068,"name":"Jean-Philippe Evrard","email":"openstack@a.spamming.party","username":"evrardjp"},"state":"REVIEWER"}],"messages":[{"id":"79fdd510f7d7d083628de0c74bf85cc895ab7d09","author":{"_account_id":6816,"name":"Jesse Pretorius","email":"jesse@odyssey4.me","username":"jesse-pretorius"},"date":"2019-05-10 15:59:30.000000000","message":"Uploaded patch set 1.","accounts_in_message":[],"_revision_number":1},{"id":"a305f4af4e02b2385b95fe2f21a4e7b029d7ec20","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2019-05-10 16:05:03.000000000","message":"Patch Set 1: Verified-1\n\nThis change depends on a change that failed to merge.","accounts_in_message":[],"_revision_number":1},{"id":"ea3182f481f26287b0d9a37fcc482c73aa0d3237","author":{"_account_id":6816,"name":"Jesse Pretorius","email":"jesse@odyssey4.me","username":"jesse-pretorius"},"date":"2019-05-10 16:05:33.000000000","message":"Patch Set 1:\n\nrecheck","accounts_in_message":[],"_revision_number":1},{"id":"adf4b6baba2bb3a97bfc0a9b76c465425aae42eb","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2019-05-10 16:18:51.000000000","message":"Patch Set 1:\n\nThis change depends on a change that failed to merge.","accounts_in_message":[],"_revision_number":1},{"id":"33a48c68e1e459538526ad3e6b74580aa1e05038","author":{"_account_id":6816,"name":"Jesse Pretorius","email":"jesse@odyssey4.me","username":"jesse-pretorius"},"date":"2019-05-10 16:19:28.000000000","message":"Patch Set 1:\n\nrecheck","accounts_in_message":[],"_revision_number":1},{"id":"ad76a33e2199f6881dfa0fc4634832f8f91cdda7","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2019-05-10 16:24:17.000000000","message":"Patch Set 1:\n\nThis change depends on a change that failed to merge.","accounts_in_message":[],"_revision_number":1},{"id":"15edb67c62ea012b9926759ca160c6ff5004b28b","author":{"_account_id":6816,"name":"Jesse Pretorius","email":"jesse@odyssey4.me","username":"jesse-pretorius"},"date":"2019-05-10 16:27:10.000000000","message":"Patch Set 1:\n\nrecheck","accounts_in_message":[],"_revision_number":1},{"id":"73de213262d3966fcf5073ca48c946ef6cdda925","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2019-05-10 16:45:26.000000000","message":"Patch Set 1:\n\nThis change depends on a change that failed to merge.","accounts_in_message":[],"_revision_number":1},{"id":"fc13270ae8d797f56ed160daedf9cd4f0d42cbe0","author":{"_account_id":6816,"name":"Jesse Pretorius","email":"jesse@odyssey4.me","username":"jesse-pretorius"},"date":"2019-05-10 16:47:08.000000000","message":"Patch Set 1:\n\nrecheck","accounts_in_message":[],"_revision_number":1},{"id":"e9d0f384e19ec53ae733484260a45e352a485908","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2019-05-10 16:58:05.000000000","message":"Patch Set 1:\n\nBuild failed (check pipeline).  For information on how to proceed, see\nhttp://docs.openstack.org/infra/manual/developers.html#automated-testing\n\n\n- openstack-ansible-linters http://logs.openstack.org/05/658405/1/check/openstack-ansible-linters/202da7e/ : SUCCESS in 4m 50s\n- openstack-ansible-deploy-aio_metal-debian-stable http://logs.openstack.org/05/658405/1/check/openstack-ansible-deploy-aio_metal-debian-stable/a69b07d/ : FAILURE in 1m 45s\n- openstack-ansible-deploy-aio_metal-centos-7 http://logs.openstack.org/05/658405/1/check/openstack-ansible-deploy-aio_metal-centos-7/d2ef8f5/ : FAILURE in 1m 43s\n- openstack-ansible-deploy-aio_metal-opensuse-150 http://logs.openstack.org/05/658405/1/check/openstack-ansible-deploy-aio_metal-opensuse-150/4978d2d/ : FAILURE in 2m 10s\n- openstack-ansible-deploy-aio_metal-ubuntu-bionic http://logs.openstack.org/05/658405/1/check/openstack-ansible-deploy-aio_metal-ubuntu-bionic/c04bb53/ : FAILURE in 1m 42s\n- openstack-tox-docs http://logs.openstack.org/05/658405/1/check/openstack-tox-docs/e07eacc/html/ : SUCCESS in 3m 49s","accounts_in_message":[],"_revision_number":1},{"id":"670cd50d1057964e9e12d1e28acd878506aee7e7","author":{"_account_id":6816,"name":"Jesse Pretorius","email":"jesse@odyssey4.me","username":"jesse-pretorius"},"date":"2019-05-10 17:19:27.000000000","message":"Patch Set 1:\n\nrecheck","accounts_in_message":[],"_revision_number":1},{"id":"c19cd8497558cecee6e1f9ecc8344e389521f099","author":{"_account_id":6816,"name":"Jesse Pretorius","email":"jesse@odyssey4.me","username":"jesse-pretorius"},"date":"2019-05-10 17:42:36.000000000","message":"Patch Set 2: Commit message was updated.","accounts_in_message":[],"_revision_number":2},{"id":"ee27c3d5da0004de46fc3e8d496872d6b5a0a66d","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2019-05-10 18:02:16.000000000","message":"Patch Set 2: Verified-1\n\nThis change depends on a change that failed to merge.","accounts_in_message":[],"_revision_number":2},{"id":"a1df3604f779c1f7b55bfc4d63d95934eda9ef5c","author":{"_account_id":6816,"name":"Jesse Pretorius","email":"jesse@odyssey4.me","username":"jesse-pretorius"},"date":"2019-05-10 18:04:16.000000000","message":"Patch Set 2:\n\nrecheck","accounts_in_message":[],"_revision_number":2},{"id":"ad41ff03a7ad14fa026c1d853e3bc6291adcd528","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2019-05-10 18:07:32.000000000","message":"Patch Set 2:\n\nThis change depends on a change that failed to merge.","accounts_in_message":[],"_revision_number":2},{"id":"58b3e51b8ef72c4376ebc2734e4d371cc417670a","author":{"_account_id":6816,"name":"Jesse Pretorius","email":"jesse@odyssey4.me","username":"jesse-pretorius"},"date":"2019-05-10 18:10:04.000000000","message":"Patch Set 2:\n\nrecheck","accounts_in_message":[],"_revision_number":2},{"id":"269c798307bc004b96354a64fefd7c771a16e867","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2019-05-10 18:14:58.000000000","message":"Patch Set 2:\n\nThis change depends on a change that failed to merge.","accounts_in_message":[],"_revision_number":2},{"id":"eea0d4b2c064bb8ad4e09b797dda3741a5b54cbb","author":{"_account_id":6816,"name":"Jesse Pretorius","email":"jesse@odyssey4.me","username":"jesse-pretorius"},"date":"2019-05-10 18:19:22.000000000","message":"Patch Set 2:\n\nrecheck","accounts_in_message":[],"_revision_number":2},{"id":"acfc3bfbe617d901490e51a9f6c07f7a20c83ca5","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2019-05-10 18:33:35.000000000","message":"Patch Set 2:\n\nBuild failed (check pipeline).  For information on how to proceed, see\nhttp://docs.openstack.org/infra/manual/developers.html#automated-testing\n\n\n- openstack-ansible-linters http://logs.openstack.org/05/658405/2/check/openstack-ansible-linters/f80ed4d/ : SUCCESS in 6m 00s\n- openstack-ansible-deploy-aio_metal-debian-stable http://logs.openstack.org/05/658405/2/check/openstack-ansible-deploy-aio_metal-debian-stable/beeddc8/ : FAILURE in 1m 59s\n- openstack-ansible-deploy-aio_metal-centos-7 http://logs.openstack.org/05/658405/2/check/openstack-ansible-deploy-aio_metal-centos-7/24c31a0/ : FAILURE in 2m 21s\n- openstack-ansible-deploy-aio_metal-opensuse-150 http://logs.openstack.org/05/658405/2/check/openstack-ansible-deploy-aio_metal-opensuse-150/0cc6f97/ : FAILURE in 2m 06s\n- openstack-ansible-deploy-aio_metal-ubuntu-bionic http://logs.openstack.org/05/658405/2/check/openstack-ansible-deploy-aio_metal-ubuntu-bionic/4f3710c/ : FAILURE in 1m 50s\n- openstack-tox-docs http://logs.openstack.org/05/658405/2/check/openstack-tox-docs/5d1e947/html/ : SUCCESS in 2m 19s","accounts_in_message":[],"_revision_number":2},{"id":"c1c61eb6fae4c3f3a6ab43b74e86be3f66201521","author":{"_account_id":6816,"name":"Jesse Pretorius","email":"jesse@odyssey4.me","username":"jesse-pretorius"},"date":"2019-05-10 19:26:40.000000000","message":"Patch Set 2:\n\nrecheck","accounts_in_message":[],"_revision_number":2},{"id":"e62fc1b642b2ca3f7bd6241e6e1fb8e42dc60b74","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2019-05-10 21:16:37.000000000","message":"Patch Set 2:\n\nBuild failed (check pipeline).  For information on how to proceed, see\nhttp://docs.openstack.org/infra/manual/developers.html#automated-testing\n\n\n- openstack-ansible-linters http://logs.openstack.org/05/658405/2/check/openstack-ansible-linters/99a7fe2/ : SUCCESS in 5m 28s\n- openstack-ansible-deploy-aio_metal-debian-stable http://logs.openstack.org/05/658405/2/check/openstack-ansible-deploy-aio_metal-debian-stable/fe6d125/ : POST_FAILURE in 9m 39s\n- openstack-ansible-deploy-aio_metal-centos-7 http://logs.openstack.org/05/658405/2/check/openstack-ansible-deploy-aio_metal-centos-7/7114832/ : SUCCESS in 1h 41m 33s\n- openstack-ansible-deploy-aio_metal-opensuse-150 http://logs.openstack.org/05/658405/2/check/openstack-ansible-deploy-aio_metal-opensuse-150/7d74be4/ : SUCCESS in 1h 22m 57s\n- openstack-ansible-deploy-aio_metal-ubuntu-bionic http://logs.openstack.org/05/658405/2/check/openstack-ansible-deploy-aio_metal-ubuntu-bionic/b0dcf85/ : SUCCESS in 1h 09m 37s\n- openstack-tox-docs http://logs.openstack.org/05/658405/2/check/openstack-tox-docs/5fe6b88/html/ : SUCCESS in 2m 50s","accounts_in_message":[],"_revision_number":2},{"id":"438886d2b48b8303a55bd69da1a5b7802c6184c9","author":{"_account_id":17068,"name":"Jean-Philippe Evrard","email":"openstack@a.spamming.party","username":"evrardjp"},"date":"2019-05-14 09:53:42.000000000","message":"Patch Set 2: Code-Review-1\n\nCan we move to this please?\nhttps://review.opendev.org/#/c/658376/ I will separate the suse meta change.","accounts_in_message":[],"_revision_number":2},{"id":"2358bd2f83f57cab12eccbe9e170c31269e6aafb","author":{"_account_id":6816,"name":"Jesse Pretorius","email":"jesse@odyssey4.me","username":"jesse-pretorius"},"date":"2019-08-07 15:38:07.000000000","message":"Abandoned\n\nNo longer working on this. Please restore this patch if it is still required.","accounts_in_message":[],"_revision_number":2}],"current_revision_number":2,"current_revision":"cf02fe8a6347db8bc353667b0ed303fadde181c5","revisions":{"0ab61b6ceecc1ac3a4fcbe78ecc0f10587633d38":{"kind":"REWORK","_number":1,"created":"2019-05-10 15:59:30.000000000","uploader":{"_account_id":6816,"name":"Jesse Pretorius","email":"jesse@odyssey4.me","username":"jesse-pretorius"},"ref":"refs/changes/05/658405/1","fetch":{"anonymous http":{"url":"https://review.opendev.org/openstack/openstack-ansible-os_aodh","ref":"refs/changes/05/658405/1","commands":{"Checkout":"git fetch https://review.opendev.org/openstack/openstack-ansible-os_aodh refs/changes/05/658405/1 \u0026\u0026 git checkout FETCH_HEAD","Cherry Pick":"git fetch https://review.opendev.org/openstack/openstack-ansible-os_aodh refs/changes/05/658405/1 \u0026\u0026 git cherry-pick FETCH_HEAD","Format Patch":"git fetch https://review.opendev.org/openstack/openstack-ansible-os_aodh refs/changes/05/658405/1 \u0026\u0026 git format-patch -1 --stdout FETCH_HEAD","Pull":"git pull https://review.opendev.org/openstack/openstack-ansible-os_aodh refs/changes/05/658405/1"}}},"commit":{"parents":[{"commit":"07ea68ebfa170c187ed9320e596ef744d7ac8277","subject":"Updated from OpenStack Ansible Tests","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/openstack-ansible-os_aodh/commit/07ea68ebfa170c187ed9320e596ef744d7ac8277"}]}],"author":{"name":"Jesse Pretorius","email":"jesse@odyssey4.me","date":"2019-05-10 15:56:55.000000000","tz":60},"committer":{"name":"Jesse Pretorius","email":"jesse@odyssey4.me","date":"2019-05-10 15:59:11.000000000","tz":60},"subject":"[TEST] Use the dynamic aio scenario","message":"[TEST] Use the dynamic aio scenario\n\nDepends-On: https://review.opendev.org/658398\nChange-Id: If693c0b5195d21414dc2217c0c04cb3bfe11049e\n","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/openstack-ansible-os_aodh/commit/0ab61b6ceecc1ac3a4fcbe78ecc0f10587633d38"}],"resolve_conflicts_web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/openstack-ansible-os_aodh/commit/0ab61b6ceecc1ac3a4fcbe78ecc0f10587633d38"}]},"branch":"refs/heads/master"},"cf02fe8a6347db8bc353667b0ed303fadde181c5":{"kind":"NO_CODE_CHANGE","_number":2,"created":"2019-05-10 17:42:36.000000000","uploader":{"_account_id":6816,"name":"Jesse Pretorius","email":"jesse@odyssey4.me","username":"jesse-pretorius"},"ref":"refs/changes/05/658405/2","fetch":{"anonymous http":{"url":"https://review.opendev.org/openstack/openstack-ansible-os_aodh","ref":"refs/changes/05/658405/2","commands":{"Checkout":"git fetch https://review.opendev.org/openstack/openstack-ansible-os_aodh refs/changes/05/658405/2 \u0026\u0026 git checkout FETCH_HEAD","Cherry Pick":"git fetch https://review.opendev.org/openstack/openstack-ansible-os_aodh refs/changes/05/658405/2 \u0026\u0026 git cherry-pick FETCH_HEAD","Format Patch":"git fetch https://review.opendev.org/openstack/openstack-ansible-os_aodh refs/changes/05/658405/2 \u0026\u0026 git format-patch -1 --stdout FETCH_HEAD","Pull":"git pull https://review.opendev.org/openstack/openstack-ansible-os_aodh refs/changes/05/658405/2"}}},"commit":{"parents":[{"commit":"07ea68ebfa170c187ed9320e596ef744d7ac8277","subject":"Updated from OpenStack Ansible Tests","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/openstack-ansible-os_aodh/commit/07ea68ebfa170c187ed9320e596ef744d7ac8277"}]}],"author":{"name":"Jesse Pretorius","email":"jesse@odyssey4.me","date":"2019-05-10 15:56:55.000000000","tz":60},"committer":{"name":"Jesse Pretorius (odyssey4me)","email":"jesse@odyssey4.me","date":"2019-05-10 17:42:33.000000000","tz":0},"subject":"Use the dynamic aio scenario","message":"Use the dynamic aio scenario\n\nRather than have to define a job per non-aio service, we have\nzuul introspect the origin test repo and add the service name\nto the scenario variable on the fly. This allows us to simplify\nour job definitions and simply use the AIO job template for all\nrepositories.\n\nDepends-On: https://review.opendev.org/658398\nChange-Id: If693c0b5195d21414dc2217c0c04cb3bfe11049e\n","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/openstack-ansible-os_aodh/commit/cf02fe8a6347db8bc353667b0ed303fadde181c5"}],"resolve_conflicts_web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/openstack-ansible-os_aodh/commit/cf02fe8a6347db8bc353667b0ed303fadde181c5"}]},"branch":"refs/heads/master"}},"requirements":[],"submit_records":[],"submit_requirements":[]}
