)]}'
{"id":"openstack%2Fopenstack-ansible-os_aodh~658586","triplet_id":"openstack%2Fopenstack-ansible-os_aodh~stable%2Fstein~I17f150fda80c20caab6111d439f21133e2835812","project":"openstack/openstack-ansible-os_aodh","branch":"stable/stein","topic":"openstack/openstack-ansible-tests/sync-tests","hashtags":[],"change_id":"I17f150fda80c20caab6111d439f21133e2835812","subject":"Updated from OpenStack Ansible Tests","status":"MERGED","created":"2019-05-11 23:58:13.000000000","updated":"2019-05-12 03:38:54.000000000","submitted":"2019-05-12 03:38:54.000000000","submitter":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"total_comment_count":0,"unresolved_comment_count":0,"has_review_started":true,"submission_id":"658586-1557632334559-6085ff31","meta_rev_id":"0c47cee30cdf7db609fb2fa8c71424ed3816b7ba","_number":658586,"virtual_id_number":658586,"owner":{"_account_id":11131,"name":"OpenStack Proposal Bot","username":"proposal-bot","tags":["SERVICE_USER"]},"actions":{},"labels":{"Verified":{"approved":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"all":[{"value":0,"_account_id":17799,"name":"Logan V","email":"logan2211@gmail.com","username":"Logan2211"},{"value":2,"date":"2019-05-12 03:38:54.000000000","permitted_voting_range":{"min":2,"max":2},"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]}],"values":{"-2":"Fails","-1":"Doesn\u0027t seem to work"," 0":"No score","+1":"Works for me","+2":"Verified"},"description":"","default_value":0,"optional":true},"Code-Review":{"approved":{"_account_id":17799,"name":"Logan V","email":"logan2211@gmail.com","username":"Logan2211"},"all":[{"value":2,"date":"2019-05-12 01:56:02.000000000","permitted_voting_range":{"min":2,"max":2},"_account_id":17799,"name":"Logan V","email":"logan2211@gmail.com","username":"Logan2211"},{"value":0,"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]}],"values":{"-2":"Do not merge","-1":"This patch needs further work before it can be merged"," 0":"No score","+1":"Looks good to me, but someone else must approve","+2":"Looks good to me (core reviewer)"},"description":"","default_value":0,"optional":true},"Workflow":{"approved":{"_account_id":17799,"name":"Logan V","email":"logan2211@gmail.com","username":"Logan2211"},"all":[{"value":1,"date":"2019-05-12 01:56:02.000000000","permitted_voting_range":{"min":1,"max":1},"_account_id":17799,"name":"Logan V","email":"logan2211@gmail.com","username":"Logan2211"},{"value":0,"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]}],"values":{"-1":"Work in progress"," 0":"Ready for reviews","+1":"Approved"},"description":"","default_value":0,"optional":true},"Backport-Candidate":{"all":[{"value":0,"_account_id":17799,"name":"Logan V","email":"logan2211@gmail.com","username":"Logan2211"},{"value":0,"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]}],"values":{"-1":"Not A Backport Candidate"," 0":"Backport Review Needed","+1":"Proposed Backport"},"description":"","default_value":0,"optional":true}},"removable_reviewers":[],"reviewers":{"REVIEWER":[{"_account_id":17799,"name":"Logan V","email":"logan2211@gmail.com","username":"Logan2211"},{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]}]},"pending_reviewers":{},"reviewer_updates":[{"updated":"2019-05-12 01:56:02.000000000","updated_by":{"_account_id":17799,"name":"Logan V","email":"logan2211@gmail.com","username":"Logan2211"},"reviewer":{"_account_id":17799,"name":"Logan V","email":"logan2211@gmail.com","username":"Logan2211"},"state":"REVIEWER"},{"updated":"2019-05-12 03:38:54.000000000","updated_by":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"reviewer":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"state":"REVIEWER"}],"messages":[{"id":"a8eccc3f2a4d9beeeafc4c4c65b9f423d45820f0","author":{"_account_id":11131,"name":"OpenStack Proposal Bot","username":"proposal-bot","tags":["SERVICE_USER"]},"date":"2019-05-11 23:58:13.000000000","message":"Uploaded patch set 1.","accounts_in_message":[],"_revision_number":1},{"id":"c1c692eac07387c30a8dbf4f5e65a90f2c85fc4f","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2019-05-12 01:37:54.000000000","message":"Patch Set 1: Verified+1\n\nBuild succeeded (check pipeline).\n\n- openstack-ansible-linters http://logs.openstack.org/86/658586/1/check/openstack-ansible-linters/cbb1156/ : SUCCESS in 5m 56s\n- openstack-ansible-deploy-aio_telemetry_metal-centos-7 http://logs.openstack.org/86/658586/1/check/openstack-ansible-deploy-aio_telemetry_metal-centos-7/53668c0/ : SUCCESS in 1h 34m 18s\n- openstack-ansible-deploy-aio_telemetry_metal-debian-stable http://logs.openstack.org/86/658586/1/check/openstack-ansible-deploy-aio_telemetry_metal-debian-stable/151c0b3/ : SUCCESS in 1h 03m 38s\n- openstack-ansible-deploy-aio_telemetry_metal-ubuntu-bionic http://logs.openstack.org/86/658586/1/check/openstack-ansible-deploy-aio_telemetry_metal-ubuntu-bionic/95b4b92/ : SUCCESS in 1h 07m 07s\n- openstack-tox-docs http://logs.openstack.org/86/658586/1/check/openstack-tox-docs/42573d4/html/ : SUCCESS in 3m 49s\n- build-openstack-releasenotes http://logs.openstack.org/86/658586/1/check/build-openstack-releasenotes/464c0fb/html/ : SUCCESS in 2m 55s","accounts_in_message":[],"_revision_number":1},{"id":"bf1acf9bff3c4ceecb351873caecdb7c43c89dd6","author":{"_account_id":17799,"name":"Logan V","email":"logan2211@gmail.com","username":"Logan2211"},"date":"2019-05-12 01:56:02.000000000","message":"Patch Set 1: Code-Review+2 Workflow+1","accounts_in_message":[],"_revision_number":1},{"id":"d1adf481ecb7a41493bafe7ae29e0ea7c81f13fa","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2019-05-12 01:56:14.000000000","message":"Patch Set 1: -Verified\n\nStarting gate jobs.","accounts_in_message":[],"_revision_number":1},{"id":"05e80acf870289e96fd66afb65b5fcea53c01807","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2019-05-12 03:38:54.000000000","message":"Change has been successfully merged by Zuul","accounts_in_message":[],"_revision_number":1},{"id":"28cbc8d7af7fbe0ca04afaeb36242d1f405a8631","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2019-05-12 03:38:54.000000000","message":"Patch Set 1: Verified+2\n\nBuild succeeded (gate pipeline).\n\n- openstack-ansible-linters http://logs.openstack.org/86/658586/1/gate/openstack-ansible-linters/00c4428/ : SUCCESS in 5m 27s\n- openstack-ansible-deploy-aio_telemetry_metal-centos-7 http://logs.openstack.org/86/658586/1/gate/openstack-ansible-deploy-aio_telemetry_metal-centos-7/b6588dd/ : SUCCESS in 1h 32m 07s\n- openstack-ansible-deploy-aio_telemetry_metal-debian-stable http://logs.openstack.org/86/658586/1/gate/openstack-ansible-deploy-aio_telemetry_metal-debian-stable/bdd25dc/ : SUCCESS in 1h 09m 48s\n- openstack-ansible-deploy-aio_telemetry_metal-ubuntu-bionic http://logs.openstack.org/86/658586/1/gate/openstack-ansible-deploy-aio_telemetry_metal-ubuntu-bionic/3d6f940/ : SUCCESS in 1h 06m 42s\n- openstack-tox-docs http://logs.openstack.org/86/658586/1/gate/openstack-tox-docs/bc38daa/html/ : SUCCESS in 2m 37s\n- build-openstack-releasenotes http://logs.openstack.org/86/658586/1/gate/build-openstack-releasenotes/e15d055/html/ : SUCCESS in 2m 42s","accounts_in_message":[],"_revision_number":1}],"current_revision_number":1,"current_revision":"ca29ed39e0ac7bd243dd40e34b5df84ec21dc74c","revisions":{"ca29ed39e0ac7bd243dd40e34b5df84ec21dc74c":{"kind":"REWORK","_number":1,"created":"2019-05-11 23:58:13.000000000","uploader":{"_account_id":11131,"name":"OpenStack Proposal Bot","username":"proposal-bot","tags":["SERVICE_USER"]},"ref":"refs/changes/86/658586/1","fetch":{"anonymous http":{"url":"https://review.opendev.org/openstack/openstack-ansible-os_aodh","ref":"refs/changes/86/658586/1","commands":{"Checkout":"git fetch https://review.opendev.org/openstack/openstack-ansible-os_aodh refs/changes/86/658586/1 \u0026\u0026 git checkout FETCH_HEAD","Cherry Pick":"git fetch https://review.opendev.org/openstack/openstack-ansible-os_aodh refs/changes/86/658586/1 \u0026\u0026 git cherry-pick FETCH_HEAD","Format Patch":"git fetch https://review.opendev.org/openstack/openstack-ansible-os_aodh refs/changes/86/658586/1 \u0026\u0026 git format-patch -1 --stdout FETCH_HEAD","Pull":"git pull https://review.opendev.org/openstack/openstack-ansible-os_aodh refs/changes/86/658586/1"}}},"commit":{"parents":[{"commit":"0de8b3d69010ad1a86ace9dce9c6edf96ef6848c","subject":"Updated from OpenStack Ansible Tests","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/openstack-ansible-os_aodh/commit/0de8b3d69010ad1a86ace9dce9c6edf96ef6848c"}]}],"author":{"name":"OpenStack Proposal Bot","email":"openstack-infra@lists.openstack.org","date":"2019-05-11 23:58:12.000000000","tz":0},"committer":{"name":"OpenStack Proposal Bot","email":"openstack-infra@lists.openstack.org","date":"2019-05-11 23:58:12.000000000","tz":0},"subject":"Updated from OpenStack Ansible Tests","message":"Updated from OpenStack Ansible Tests\n\nChange-Id: I17f150fda80c20caab6111d439f21133e2835812\n","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/openstack-ansible-os_aodh/commit/ca29ed39e0ac7bd243dd40e34b5df84ec21dc74c"}],"resolve_conflicts_web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/openstack-ansible-os_aodh/commit/ca29ed39e0ac7bd243dd40e34b5df84ec21dc74c"}]},"branch":"refs/heads/stable/stein"}},"requirements":[],"submit_records":[],"submit_requirements":[]}
