)]}'
{"id":"openstack%2Fopenstack-ansible-os_aodh~659010","triplet_id":"openstack%2Fopenstack-ansible-os_aodh~master~Icd2d1f8f4ee930416c81b9d00fa2e3bc36dac386","project":"openstack/openstack-ansible-os_aodh","branch":"master","topic":"clarify-suse-support","hashtags":[],"change_id":"Icd2d1f8f4ee930416c81b9d00fa2e3bc36dac386","subject":"Clarify supported SUSE versions","status":"MERGED","created":"2019-05-14 09:58:00.000000000","updated":"2019-05-15 12:10:11.000000000","submitted":"2019-05-15 12:10:11.000000000","submitter":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"total_comment_count":0,"unresolved_comment_count":0,"has_review_started":true,"submission_id":"659010-1557922211334-881d6cdd","meta_rev_id":"3c0adf4b1d5ff4c10ac4ad4f2060bf497b4987ee","_number":659010,"virtual_id_number":659010,"owner":{"_account_id":17068,"name":"Jean-Philippe Evrard","email":"openstack@a.spamming.party","username":"evrardjp"},"actions":{},"labels":{"Verified":{"approved":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"all":[{"value":0,"_account_id":19298,"name":"Nicolas Bock","email":"nicolas.bock@canonical.com","username":"nicolasbock"},{"value":0,"_account_id":28935,"name":"98k","email":"18552437190@163.com","username":"98k"},{"value":0,"_account_id":28619,"name":"Dmitriy Rabotyagov","email":"noonedeadpunk@gmail.com","username":"noonedeadpunk"},{"value":0,"date":"2019-05-14 16:25:32.000000000","_account_id":17068,"name":"Jean-Philippe Evrard","email":"openstack@a.spamming.party","username":"evrardjp"},{"value":2,"date":"2019-05-15 12:10:11.000000000","permitted_voting_range":{"min":2,"max":2},"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},{"value":0,"_account_id":25023,"name":"Jonathan Rosser","email":"jonathan.rosser@rd.bbc.co.uk","username":"jrosser"}],"values":{"-2":"Fails","-1":"Doesn\u0027t seem to work"," 0":"No score","+1":"Works for me","+2":"Verified"},"description":"","default_value":0,"optional":true},"Code-Review":{"approved":{"_account_id":28619,"name":"Dmitriy Rabotyagov","email":"noonedeadpunk@gmail.com","username":"noonedeadpunk"},"recommended":{"_account_id":28935,"name":"98k","email":"18552437190@163.com","username":"98k"},"all":[{"value":1,"date":"2019-05-14 14:58:05.000000000","permitted_voting_range":{"min":1,"max":1},"_account_id":19298,"name":"Nicolas Bock","email":"nicolas.bock@canonical.com","username":"nicolasbock"},{"value":1,"date":"2019-05-14 14:51:54.000000000","permitted_voting_range":{"min":1,"max":1},"_account_id":28935,"name":"98k","email":"18552437190@163.com","username":"98k"},{"value":2,"date":"2019-05-14 13:42:34.000000000","permitted_voting_range":{"min":2,"max":2},"_account_id":28619,"name":"Dmitriy Rabotyagov","email":"noonedeadpunk@gmail.com","username":"noonedeadpunk"},{"value":0,"_account_id":17068,"name":"Jean-Philippe Evrard","email":"openstack@a.spamming.party","username":"evrardjp"},{"value":0,"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},{"value":2,"date":"2019-05-14 21:16:02.000000000","permitted_voting_range":{"min":2,"max":2},"_account_id":25023,"name":"Jonathan Rosser","email":"jonathan.rosser@rd.bbc.co.uk","username":"jrosser"}],"values":{"-2":"Do not merge","-1":"This patch needs further work before it can be merged"," 0":"No score","+1":"Looks good to me, but someone else must approve","+2":"Looks good to me (core reviewer)"},"description":"","value":1,"default_value":0,"optional":true},"Workflow":{"approved":{"_account_id":25023,"name":"Jonathan Rosser","email":"jonathan.rosser@rd.bbc.co.uk","username":"jrosser"},"all":[{"value":0,"_account_id":19298,"name":"Nicolas Bock","email":"nicolas.bock@canonical.com","username":"nicolasbock"},{"value":0,"_account_id":28935,"name":"98k","email":"18552437190@163.com","username":"98k"},{"value":0,"_account_id":28619,"name":"Dmitriy Rabotyagov","email":"noonedeadpunk@gmail.com","username":"noonedeadpunk"},{"value":0,"_account_id":17068,"name":"Jean-Philippe Evrard","email":"openstack@a.spamming.party","username":"evrardjp"},{"value":0,"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},{"value":1,"date":"2019-05-14 21:16:02.000000000","permitted_voting_range":{"min":1,"max":1},"_account_id":25023,"name":"Jonathan Rosser","email":"jonathan.rosser@rd.bbc.co.uk","username":"jrosser"}],"values":{"-1":"Work in progress"," 0":"Ready for reviews","+1":"Approved"},"description":"","default_value":0,"optional":true},"Backport-Candidate":{"all":[{"value":0,"_account_id":19298,"name":"Nicolas Bock","email":"nicolas.bock@canonical.com","username":"nicolasbock"},{"value":0,"_account_id":28935,"name":"98k","email":"18552437190@163.com","username":"98k"},{"value":0,"_account_id":28619,"name":"Dmitriy Rabotyagov","email":"noonedeadpunk@gmail.com","username":"noonedeadpunk"},{"value":0,"_account_id":17068,"name":"Jean-Philippe Evrard","email":"openstack@a.spamming.party","username":"evrardjp"},{"value":0,"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},{"value":0,"_account_id":25023,"name":"Jonathan Rosser","email":"jonathan.rosser@rd.bbc.co.uk","username":"jrosser"}],"values":{"-1":"Not A Backport Candidate"," 0":"Backport Review Needed","+1":"Proposed Backport"},"description":"","default_value":0,"optional":true}},"removable_reviewers":[],"reviewers":{"REVIEWER":[{"_account_id":17068,"name":"Jean-Philippe Evrard","email":"openstack@a.spamming.party","username":"evrardjp"},{"_account_id":19298,"name":"Nicolas Bock","email":"nicolas.bock@canonical.com","username":"nicolasbock"},{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},{"_account_id":25023,"name":"Jonathan Rosser","email":"jonathan.rosser@rd.bbc.co.uk","username":"jrosser"},{"_account_id":28619,"name":"Dmitriy Rabotyagov","email":"noonedeadpunk@gmail.com","username":"noonedeadpunk"},{"_account_id":28935,"name":"98k","email":"18552437190@163.com","username":"98k"}]},"pending_reviewers":{},"reviewer_updates":[{"updated":"2019-05-14 10:02:49.000000000","updated_by":{"_account_id":28619,"name":"Dmitriy Rabotyagov","email":"noonedeadpunk@gmail.com","username":"noonedeadpunk"},"reviewer":{"_account_id":28619,"name":"Dmitriy Rabotyagov","email":"noonedeadpunk@gmail.com","username":"noonedeadpunk"},"state":"REVIEWER"},{"updated":"2019-05-14 14:51:54.000000000","updated_by":{"_account_id":28935,"name":"98k","email":"18552437190@163.com","username":"98k"},"reviewer":{"_account_id":28935,"name":"98k","email":"18552437190@163.com","username":"98k"},"state":"REVIEWER"},{"updated":"2019-05-14 14:58:05.000000000","updated_by":{"_account_id":19298,"name":"Nicolas Bock","email":"nicolas.bock@canonical.com","username":"nicolasbock"},"reviewer":{"_account_id":19298,"name":"Nicolas Bock","email":"nicolas.bock@canonical.com","username":"nicolasbock"},"state":"REVIEWER"},{"updated":"2019-05-14 21:16:02.000000000","updated_by":{"_account_id":25023,"name":"Jonathan Rosser","email":"jonathan.rosser@rd.bbc.co.uk","username":"jrosser"},"reviewer":{"_account_id":25023,"name":"Jonathan Rosser","email":"jonathan.rosser@rd.bbc.co.uk","username":"jrosser"},"state":"REVIEWER"},{"updated":"2019-05-15 12:10:11.000000000","updated_by":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"reviewer":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"state":"REVIEWER"}],"messages":[{"id":"145eb0af5b2abb62484ffd44c0dceda8908f79a3","author":{"_account_id":17068,"name":"Jean-Philippe Evrard","email":"openstack@a.spamming.party","username":"evrardjp"},"date":"2019-05-14 09:58:00.000000000","message":"Uploaded patch set 1.","accounts_in_message":[],"_revision_number":1},{"id":"4af020b74dd180c77ac1421a5f42899e61f4c7a0","author":{"_account_id":17068,"name":"Jean-Philippe Evrard","email":"openstack@a.spamming.party","username":"evrardjp"},"date":"2019-05-14 09:58:14.000000000","message":"Topic changed from use-integrated-repo-for-all-tests to clarify-suse-support","accounts_in_message":[],"_revision_number":1},{"id":"080b00dfb9d21fbf60196853e1a1c4395949e507","author":{"_account_id":28619,"name":"Dmitriy Rabotyagov","email":"noonedeadpunk@gmail.com","username":"noonedeadpunk"},"date":"2019-05-14 10:02:49.000000000","message":"Patch Set 1: Code-Review+2","accounts_in_message":[],"_revision_number":1},{"id":"db6d427c2089d6d1c9d9124f5a5862daa709147e","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2019-05-14 10:52:26.000000000","message":"Patch Set 1: Verified-1\n\nBuild failed (check pipeline).  For information on how to proceed, see\nhttp://docs.openstack.org/infra/manual/developers.html#automated-testing\n\n\n- openstack-ansible-linters http://logs.openstack.org/10/659010/1/check/openstack-ansible-linters/02a0b8d/ : SUCCESS in 5m 04s\n- openstack-ansible-deploy-aio_metal-debian-stable http://logs.openstack.org/10/659010/1/check/openstack-ansible-deploy-aio_metal-debian-stable/cb45820/ : RETRY_LIMIT in 2m 02s\n- openstack-ansible-deploy-aio_metal-centos-7 http://logs.openstack.org/10/659010/1/check/openstack-ansible-deploy-aio_metal-centos-7/db81ef4/ : RETRY_LIMIT in 2m 00s\n- openstack-ansible-deploy-aio_metal-opensuse-150 http://logs.openstack.org/10/659010/1/check/openstack-ansible-deploy-aio_metal-opensuse-150/30dc98d/ : RETRY_LIMIT in 2m 10s\n- openstack-ansible-deploy-aio_metal-ubuntu-bionic http://logs.openstack.org/10/659010/1/check/openstack-ansible-deploy-aio_metal-ubuntu-bionic/fc7cf99/ : RETRY_LIMIT in 1m 51s\n- openstack-ansible-deploy-aio_distro_metal-centos-7 http://logs.openstack.org/10/659010/1/check/openstack-ansible-deploy-aio_distro_metal-centos-7/8b23106/ : RETRY_LIMIT in 2m 26s (non-voting)\n- openstack-ansible-deploy-aio_telemetry_metal-centos-7 http://logs.openstack.org/10/659010/1/check/openstack-ansible-deploy-aio_telemetry_metal-centos-7/fe52c1d/ : RETRY_LIMIT in 2m 09s\n- openstack-ansible-deploy-aio_telemetry_metal-debian-stable http://logs.openstack.org/10/659010/1/check/openstack-ansible-deploy-aio_telemetry_metal-debian-stable/037589f/ : RETRY_LIMIT in 1m 51s\n- openstack-ansible-deploy-aio_telemetry_metal-ubuntu-bionic http://logs.openstack.org/10/659010/1/check/openstack-ansible-deploy-aio_telemetry_metal-ubuntu-bionic/e49f1e4/ : RETRY_LIMIT in 1m 48s\n- openstack-tox-docs http://logs.openstack.org/10/659010/1/check/openstack-tox-docs/0e103a3/html/ : SUCCESS in 2m 50s","accounts_in_message":[],"_revision_number":1},{"id":"b406797e5d6b1f662b58c73829f080d8840a42d9","author":{"_account_id":28619,"name":"Dmitriy Rabotyagov","email":"noonedeadpunk@gmail.com","username":"noonedeadpunk"},"date":"2019-05-14 13:13:32.000000000","message":"Patch Set 1:\n\nrecheck","accounts_in_message":[],"_revision_number":1},{"id":"a85f7630f5ad866cf65f1f26393cc198002c4a25","author":{"_account_id":28619,"name":"Dmitriy Rabotyagov","email":"noonedeadpunk@gmail.com","username":"noonedeadpunk"},"date":"2019-05-14 13:42:34.000000000","message":"Patch Set 2: Patch Set 1 was rebased","accounts_in_message":[],"_revision_number":2},{"id":"e53fd84ed30335fc6d3f0eb4a46693c78e88af7d","author":{"_account_id":28935,"name":"98k","email":"18552437190@163.com","username":"98k"},"date":"2019-05-14 14:51:54.000000000","message":"Patch Set 2: Code-Review+1","accounts_in_message":[],"_revision_number":2},{"id":"334b3f53e6886381ed07cb14b2fc4ee9014ba6a4","author":{"_account_id":19298,"name":"Nicolas Bock","email":"nicolas.bock@canonical.com","username":"nicolasbock"},"date":"2019-05-14 14:58:05.000000000","message":"Patch Set 2: Code-Review+1\n\nIt\u0027s not _that_ confusing ;)","accounts_in_message":[],"_revision_number":2},{"id":"128bd8ad07f1cd10f35686ade5ee94b6b154d384","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2019-05-14 15:29:16.000000000","message":"Patch Set 2:\n\nThis change depends on a change that failed to merge.","accounts_in_message":[],"_revision_number":2},{"id":"b2758a8b77b76560c263cb666de0ccd76941998f","author":{"_account_id":17068,"name":"Jean-Philippe Evrard","email":"openstack@a.spamming.party","username":"evrardjp"},"date":"2019-05-14 16:25:12.000000000","message":"Patch Set 2:\n\nrecheck","accounts_in_message":[],"_revision_number":2},{"id":"c3fdcc2866406b2ea19cac3bef487473d53e818d","author":{"_account_id":17068,"name":"Jean-Philippe Evrard","email":"openstack@a.spamming.party","username":"evrardjp"},"date":"2019-05-14 16:25:32.000000000","message":"Patch Set 2:\n\n\u003e It\u0027s not _that_ confusing ;)\n\n:)","accounts_in_message":[],"_revision_number":2},{"id":"95b418215e2ee13edac5f4716a07302244c2962c","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2019-05-14 18:27:27.000000000","message":"Patch Set 2: Verified+1\n\nBuild succeeded (check pipeline).\n\n- openstack-ansible-linters http://logs.openstack.org/10/659010/2/check/openstack-ansible-linters/f5d02e5/ : SUCCESS in 8m 33s\n- openstack-ansible-deploy-aio_metal-debian-stable http://logs.openstack.org/10/659010/2/check/openstack-ansible-deploy-aio_metal-debian-stable/de80404/ : SUCCESS in 1h 11m 11s\n- openstack-ansible-deploy-aio_metal-centos-7 http://logs.openstack.org/10/659010/2/check/openstack-ansible-deploy-aio_metal-centos-7/56ed13b/ : SUCCESS in 1h 33m 05s\n- openstack-ansible-deploy-aio_metal-opensuse-150 http://logs.openstack.org/10/659010/2/check/openstack-ansible-deploy-aio_metal-opensuse-150/80ee5cc/ : SUCCESS in 1h 25m 38s\n- openstack-ansible-deploy-aio_metal-ubuntu-bionic http://logs.openstack.org/10/659010/2/check/openstack-ansible-deploy-aio_metal-ubuntu-bionic/41236a6/ : SUCCESS in 1h 11m 45s\n- openstack-ansible-deploy-aio_distro_metal-centos-7 http://logs.openstack.org/10/659010/2/check/openstack-ansible-deploy-aio_distro_metal-centos-7/d01abea/ : FAILURE in 42m 10s (non-voting)\n- openstack-ansible-deploy-aio_telemetry_metal-centos-7 http://logs.openstack.org/10/659010/2/check/openstack-ansible-deploy-aio_telemetry_metal-centos-7/2273870/ : SUCCESS in 1h 40m 15s\n- openstack-ansible-deploy-aio_telemetry_metal-debian-stable http://logs.openstack.org/10/659010/2/check/openstack-ansible-deploy-aio_telemetry_metal-debian-stable/69bf2dd/ : SUCCESS in 1h 07m 29s\n- openstack-ansible-deploy-aio_telemetry_metal-ubuntu-bionic http://logs.openstack.org/10/659010/2/check/openstack-ansible-deploy-aio_telemetry_metal-ubuntu-bionic/00aaced/ : SUCCESS in 1h 21m 41s\n- openstack-tox-docs http://logs.openstack.org/10/659010/2/check/openstack-tox-docs/b41c24b/html/ : SUCCESS in 3m 21s","accounts_in_message":[],"_revision_number":2},{"id":"4a4a9629098c77dac609c699906ff75860e45050","author":{"_account_id":25023,"name":"Jonathan Rosser","email":"jonathan.rosser@rd.bbc.co.uk","username":"jrosser"},"date":"2019-05-14 21:16:02.000000000","message":"Patch Set 2: Code-Review+2 Workflow+1","accounts_in_message":[],"_revision_number":2},{"id":"b575e577c2ede7cd02071f2250429871633eeee2","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2019-05-15 09:50:12.000000000","message":"Patch Set 2: -Verified\n\nStarting gate jobs.","accounts_in_message":[],"_revision_number":2},{"id":"6c9506c0e9000066891dd2bb9b0397afb93ecb84","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2019-05-15 12:10:11.000000000","message":"Change has been successfully merged by Zuul","accounts_in_message":[],"_revision_number":2},{"id":"3c0adf4b1d5ff4c10ac4ad4f2060bf497b4987ee","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2019-05-15 12:10:11.000000000","message":"Patch Set 2: Verified+2\n\nBuild succeeded (gate pipeline).\n\n- openstack-ansible-linters http://logs.openstack.org/10/659010/2/gate/openstack-ansible-linters/1286945/ : SUCCESS in 5m 02s\n- openstack-ansible-deploy-aio_metal-debian-stable http://logs.openstack.org/10/659010/2/gate/openstack-ansible-deploy-aio_metal-debian-stable/a2cec88/ : SUCCESS in 1h 21m 05s\n- openstack-ansible-deploy-aio_metal-centos-7 http://logs.openstack.org/10/659010/2/gate/openstack-ansible-deploy-aio_metal-centos-7/b716e13/ : SUCCESS in 1h 32m 41s\n- openstack-ansible-deploy-aio_metal-opensuse-150 http://logs.openstack.org/10/659010/2/gate/openstack-ansible-deploy-aio_metal-opensuse-150/a97a459/ : SUCCESS in 1h 26m 36s\n- openstack-ansible-deploy-aio_metal-ubuntu-bionic http://logs.openstack.org/10/659010/2/gate/openstack-ansible-deploy-aio_metal-ubuntu-bionic/97ecc02/ : SUCCESS in 1h 09m 07s\n- openstack-ansible-deploy-aio_telemetry_metal-centos-7 http://logs.openstack.org/10/659010/2/gate/openstack-ansible-deploy-aio_telemetry_metal-centos-7/2c3a372/ : SUCCESS in 1h 32m 19s\n- openstack-ansible-deploy-aio_telemetry_metal-debian-stable http://logs.openstack.org/10/659010/2/gate/openstack-ansible-deploy-aio_telemetry_metal-debian-stable/69bb4a0/ : SUCCESS in 1h 14m 37s\n- openstack-ansible-deploy-aio_telemetry_metal-ubuntu-bionic http://logs.openstack.org/10/659010/2/gate/openstack-ansible-deploy-aio_telemetry_metal-ubuntu-bionic/e828b32/ : SUCCESS in 1h 04m 00s\n- openstack-tox-docs http://logs.openstack.org/10/659010/2/gate/openstack-tox-docs/122df4c/html/ : SUCCESS in 2m 16s","accounts_in_message":[],"_revision_number":2}],"current_revision_number":2,"current_revision":"381061488d115283eccaf6157f4ff28ec5e699a0","revisions":{"c0f3afb92575ee20af71e1674382a7f2145ed3c7":{"kind":"REWORK","_number":1,"created":"2019-05-14 09:58:00.000000000","uploader":{"_account_id":17068,"name":"Jean-Philippe Evrard","email":"openstack@a.spamming.party","username":"evrardjp"},"ref":"refs/changes/10/659010/1","fetch":{"anonymous http":{"url":"https://review.opendev.org/openstack/openstack-ansible-os_aodh","ref":"refs/changes/10/659010/1","commands":{"Checkout":"git fetch https://review.opendev.org/openstack/openstack-ansible-os_aodh refs/changes/10/659010/1 \u0026\u0026 git checkout FETCH_HEAD","Cherry Pick":"git fetch https://review.opendev.org/openstack/openstack-ansible-os_aodh refs/changes/10/659010/1 \u0026\u0026 git cherry-pick FETCH_HEAD","Format Patch":"git fetch https://review.opendev.org/openstack/openstack-ansible-os_aodh refs/changes/10/659010/1 \u0026\u0026 git format-patch -1 --stdout FETCH_HEAD","Pull":"git pull https://review.opendev.org/openstack/openstack-ansible-os_aodh refs/changes/10/659010/1"}}},"commit":{"parents":[{"commit":"fe17d08fda14aa5585f95c4ab8f3cc9018b99ebb","subject":"Add integrated coverage","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/openstack-ansible-os_aodh/commit/fe17d08fda14aa5585f95c4ab8f3cc9018b99ebb"}]}],"author":{"name":"Jean-Philippe Evrard","email":"jean-philippe@evrard.me","date":"2019-05-14 09:55:56.000000000","tz":120},"committer":{"name":"Jean-Philippe Evrard","email":"jean-philippe@evrard.me","date":"2019-05-14 09:57:31.000000000","tz":120},"subject":"Clarify supported SUSE versions","message":"Clarify supported SUSE versions\n\nThis role only support openSUSE Leap 15, not 42.3, not tumbleweed.\nSaying all is confusing and a bad practice. This fixes it, while\nensuring the job is properly defined to test that assertion.\n\nChange-Id: Icd2d1f8f4ee930416c81b9d00fa2e3bc36dac386\n","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/openstack-ansible-os_aodh/commit/c0f3afb92575ee20af71e1674382a7f2145ed3c7"}],"resolve_conflicts_web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/openstack-ansible-os_aodh/commit/c0f3afb92575ee20af71e1674382a7f2145ed3c7"}]},"branch":"refs/heads/master"},"381061488d115283eccaf6157f4ff28ec5e699a0":{"kind":"NO_CHANGE","_number":2,"created":"2019-05-14 13:42:34.000000000","uploader":{"_account_id":28619,"name":"Dmitriy Rabotyagov","email":"noonedeadpunk@gmail.com","username":"noonedeadpunk"},"ref":"refs/changes/10/659010/2","fetch":{"anonymous http":{"url":"https://review.opendev.org/openstack/openstack-ansible-os_aodh","ref":"refs/changes/10/659010/2","commands":{"Checkout":"git fetch https://review.opendev.org/openstack/openstack-ansible-os_aodh refs/changes/10/659010/2 \u0026\u0026 git checkout FETCH_HEAD","Cherry Pick":"git fetch https://review.opendev.org/openstack/openstack-ansible-os_aodh refs/changes/10/659010/2 \u0026\u0026 git cherry-pick FETCH_HEAD","Format Patch":"git fetch https://review.opendev.org/openstack/openstack-ansible-os_aodh refs/changes/10/659010/2 \u0026\u0026 git format-patch -1 --stdout FETCH_HEAD","Pull":"git pull https://review.opendev.org/openstack/openstack-ansible-os_aodh refs/changes/10/659010/2"}}},"commit":{"parents":[{"commit":"8585f43eee839b115cafa6e15e6fc765a001db89","subject":"Add integrated coverage","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/openstack-ansible-os_aodh/commit/8585f43eee839b115cafa6e15e6fc765a001db89"}]}],"author":{"name":"Jean-Philippe Evrard","email":"jean-philippe@evrard.me","date":"2019-05-14 09:55:56.000000000","tz":120},"committer":{"name":"Dmitriy Rabotyagov (noonedeadpunk)","email":"noonedeadpunk@ya.ru","date":"2019-05-14 13:42:34.000000000","tz":0},"subject":"Clarify supported SUSE versions","message":"Clarify supported SUSE versions\n\nThis role only support openSUSE Leap 15, not 42.3, not tumbleweed.\nSaying all is confusing and a bad practice. This fixes it, while\nensuring the job is properly defined to test that assertion.\n\nChange-Id: Icd2d1f8f4ee930416c81b9d00fa2e3bc36dac386\n","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/openstack-ansible-os_aodh/commit/381061488d115283eccaf6157f4ff28ec5e699a0"}],"resolve_conflicts_web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/openstack-ansible-os_aodh/commit/381061488d115283eccaf6157f4ff28ec5e699a0"}]},"branch":"refs/heads/master"}},"requirements":[],"submit_records":[],"submit_requirements":[]}
