)]}'
{"id":"openstack%2Fopenstack-ansible-os_aodh~664303","triplet_id":"openstack%2Fopenstack-ansible-os_aodh~stable%2Frocky~I5c828c26c4cf77f66aa9dd3dcd257263b91af090","project":"openstack/openstack-ansible-os_aodh","branch":"stable/rocky","topic":"openstack/openstack-ansible-tests/sync-tests","hashtags":[],"change_id":"I5c828c26c4cf77f66aa9dd3dcd257263b91af090","subject":"Updated from OpenStack Ansible Tests","status":"MERGED","created":"2019-06-10 15:43:51.000000000","updated":"2019-06-10 18:09:12.000000000","submitted":"2019-06-10 18:09:12.000000000","submitter":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"total_comment_count":0,"unresolved_comment_count":0,"has_review_started":true,"submission_id":"664303-1560190152831-7fce86ad","meta_rev_id":"6ddadec41ebe13f48c2dde7f965ee30bf0c9e038","_number":664303,"virtual_id_number":664303,"owner":{"_account_id":11131,"name":"OpenStack Proposal Bot","username":"proposal-bot","tags":["SERVICE_USER"]},"actions":{},"labels":{"Verified":{"approved":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"all":[{"value":0,"_account_id":28619,"name":"Dmitriy Rabotyagov","email":"noonedeadpunk@gmail.com","username":"noonedeadpunk"},{"value":2,"date":"2019-06-10 18:09:12.000000000","permitted_voting_range":{"min":2,"max":2},"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]}],"values":{"-2":"Fails","-1":"Doesn\u0027t seem to work"," 0":"No score","+1":"Works for me","+2":"Verified"},"description":"","default_value":0,"optional":true},"Code-Review":{"approved":{"_account_id":28619,"name":"Dmitriy Rabotyagov","email":"noonedeadpunk@gmail.com","username":"noonedeadpunk"},"all":[{"value":2,"date":"2019-06-10 17:18:53.000000000","permitted_voting_range":{"min":2,"max":2},"_account_id":28619,"name":"Dmitriy Rabotyagov","email":"noonedeadpunk@gmail.com","username":"noonedeadpunk"},{"value":0,"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]}],"values":{"-2":"Do not merge","-1":"This patch needs further work before it can be merged"," 0":"No score","+1":"Looks good to me, but someone else must approve","+2":"Looks good to me (core reviewer)"},"description":"","default_value":0,"optional":true},"Workflow":{"approved":{"_account_id":28619,"name":"Dmitriy Rabotyagov","email":"noonedeadpunk@gmail.com","username":"noonedeadpunk"},"all":[{"value":1,"date":"2019-06-10 17:18:53.000000000","permitted_voting_range":{"min":1,"max":1},"_account_id":28619,"name":"Dmitriy Rabotyagov","email":"noonedeadpunk@gmail.com","username":"noonedeadpunk"},{"value":0,"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]}],"values":{"-1":"Work in progress"," 0":"Ready for reviews","+1":"Approved"},"description":"","default_value":0,"optional":true},"Backport-Candidate":{"all":[{"value":0,"_account_id":28619,"name":"Dmitriy Rabotyagov","email":"noonedeadpunk@gmail.com","username":"noonedeadpunk"},{"value":0,"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]}],"values":{"-1":"Not A Backport Candidate"," 0":"Backport Review Needed","+1":"Proposed Backport"},"description":"","default_value":0,"optional":true}},"removable_reviewers":[],"reviewers":{"REVIEWER":[{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},{"_account_id":28619,"name":"Dmitriy Rabotyagov","email":"noonedeadpunk@gmail.com","username":"noonedeadpunk"}]},"pending_reviewers":{},"reviewer_updates":[{"updated":"2019-06-10 17:18:53.000000000","updated_by":{"_account_id":28619,"name":"Dmitriy Rabotyagov","email":"noonedeadpunk@gmail.com","username":"noonedeadpunk"},"reviewer":{"_account_id":28619,"name":"Dmitriy Rabotyagov","email":"noonedeadpunk@gmail.com","username":"noonedeadpunk"},"state":"REVIEWER"},{"updated":"2019-06-10 18:09:12.000000000","updated_by":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"reviewer":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"state":"REVIEWER"}],"messages":[{"id":"e8c8f99bb05a851b5e7cffc4cc3142cc486165fe","author":{"_account_id":11131,"name":"OpenStack Proposal Bot","username":"proposal-bot","tags":["SERVICE_USER"]},"date":"2019-06-10 15:43:51.000000000","message":"Uploaded patch set 1.","accounts_in_message":[],"_revision_number":1},{"id":"11887c0a01acab2921463e38fe9c604229614ad7","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2019-06-10 16:43:56.000000000","message":"Patch Set 1: Verified+1\n\nBuild succeeded (check pipeline).\n\n- openstack-ansible-linters http://logs.openstack.org/03/664303/1/check/openstack-ansible-linters/eb6401d/ : SUCCESS in 5m 08s\n- openstack-ansible-functional-centos-7 http://logs.openstack.org/03/664303/1/check/openstack-ansible-functional-centos-7/2d81747/ : SUCCESS in 38m 04s\n- openstack-ansible-functional-opensuse-423 http://logs.openstack.org/03/664303/1/check/openstack-ansible-functional-opensuse-423/3b29228/ : FAILURE in 7m 38s (non-voting)\n- openstack-ansible-functional-opensuse-150 http://logs.openstack.org/03/664303/1/check/openstack-ansible-functional-opensuse-150/1b053b6/ : FAILURE in 9m 19s (non-voting)\n- openstack-ansible-functional-ubuntu-bionic http://logs.openstack.org/03/664303/1/check/openstack-ansible-functional-ubuntu-bionic/ee385bd/ : SUCCESS in 30m 03s\n- openstack-ansible-functional-ubuntu-xenial http://logs.openstack.org/03/664303/1/check/openstack-ansible-functional-ubuntu-xenial/30f7f41/ : SUCCESS in 28m 08s\n- build-openstack-sphinx-docs http://logs.openstack.org/03/664303/1/check/build-openstack-sphinx-docs/426de23/html/ : SUCCESS in 3m 35s\n- openstack-ansible-functional-distro_install-ubuntu-xenial http://logs.openstack.org/03/664303/1/check/openstack-ansible-functional-distro_install-ubuntu-xenial/c7be973/ : SUCCESS in 33m 20s\n- openstack-ansible-functional-distro_install-centos-7 http://logs.openstack.org/03/664303/1/check/openstack-ansible-functional-distro_install-centos-7/f8e5654/ : SUCCESS in 40m 00s\n- openstack-ansible-functional-distro_install-opensuse-423 http://logs.openstack.org/03/664303/1/check/openstack-ansible-functional-distro_install-opensuse-423/68d32d2/ : FAILURE in 7m 58s (non-voting)","accounts_in_message":[],"_revision_number":1},{"id":"ce965f1e5dd8d0a3952e7fc755e4e479d7a95cec","author":{"_account_id":28619,"name":"Dmitriy Rabotyagov","email":"noonedeadpunk@gmail.com","username":"noonedeadpunk"},"date":"2019-06-10 17:18:53.000000000","message":"Patch Set 1: Code-Review+2 Workflow+1","accounts_in_message":[],"_revision_number":1},{"id":"5f77d33dcf0cca9c6457e6445a51836c33377dd1","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2019-06-10 17:19:05.000000000","message":"Patch Set 1: -Verified\n\nStarting gate jobs.","accounts_in_message":[],"_revision_number":1},{"id":"752a87db35d3b01f20ba978a7c2140a5f7bdf131","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2019-06-10 18:09:12.000000000","message":"Change has been successfully merged by Zuul","accounts_in_message":[],"_revision_number":1},{"id":"e397be3f56d8922a959bd345d2a222e44d00a70c","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2019-06-10 18:09:12.000000000","message":"Patch Set 1: Verified+2\n\nBuild succeeded (gate pipeline).\n\n- openstack-ansible-linters http://logs.openstack.org/03/664303/1/gate/openstack-ansible-linters/04601a3/ : SUCCESS in 6m 58s\n- openstack-ansible-functional-centos-7 http://logs.openstack.org/03/664303/1/gate/openstack-ansible-functional-centos-7/3f923e0/ : SUCCESS in 41m 13s\n- openstack-ansible-functional-opensuse-423 http://logs.openstack.org/03/664303/1/gate/openstack-ansible-functional-opensuse-423/9553143/ : FAILURE in 9m 30s (non-voting)\n- openstack-ansible-functional-ubuntu-bionic http://logs.openstack.org/03/664303/1/gate/openstack-ansible-functional-ubuntu-bionic/0aee25e/ : SUCCESS in 37m 00s\n- openstack-ansible-functional-ubuntu-xenial http://logs.openstack.org/03/664303/1/gate/openstack-ansible-functional-ubuntu-xenial/7dd519e/ : SUCCESS in 35m 16s\n- build-openstack-sphinx-docs http://logs.openstack.org/03/664303/1/gate/build-openstack-sphinx-docs/63c6e5b/html/ : SUCCESS in 3m 35s\n- openstack-ansible-functional-distro_install-ubuntu-xenial http://logs.openstack.org/03/664303/1/gate/openstack-ansible-functional-distro_install-ubuntu-xenial/43c67f8/ : SUCCESS in 31m 31s\n- openstack-ansible-functional-distro_install-centos-7 http://logs.openstack.org/03/664303/1/gate/openstack-ansible-functional-distro_install-centos-7/f2af5cb/ : SUCCESS in 40m 06s\n- openstack-ansible-functional-distro_install-opensuse-423 http://logs.openstack.org/03/664303/1/gate/openstack-ansible-functional-distro_install-opensuse-423/61f2079/ : FAILURE in 10m 19s (non-voting)","accounts_in_message":[],"_revision_number":1}],"current_revision_number":1,"current_revision":"52a70e898b71247d59623d4559b3bda748b22be6","revisions":{"52a70e898b71247d59623d4559b3bda748b22be6":{"kind":"REWORK","_number":1,"created":"2019-06-10 15:43:51.000000000","uploader":{"_account_id":11131,"name":"OpenStack Proposal Bot","username":"proposal-bot","tags":["SERVICE_USER"]},"ref":"refs/changes/03/664303/1","fetch":{"anonymous http":{"url":"https://review.opendev.org/openstack/openstack-ansible-os_aodh","ref":"refs/changes/03/664303/1","commands":{"Checkout":"git fetch https://review.opendev.org/openstack/openstack-ansible-os_aodh refs/changes/03/664303/1 \u0026\u0026 git checkout FETCH_HEAD","Cherry Pick":"git fetch https://review.opendev.org/openstack/openstack-ansible-os_aodh refs/changes/03/664303/1 \u0026\u0026 git cherry-pick FETCH_HEAD","Format Patch":"git fetch https://review.opendev.org/openstack/openstack-ansible-os_aodh refs/changes/03/664303/1 \u0026\u0026 git format-patch -1 --stdout FETCH_HEAD","Pull":"git pull https://review.opendev.org/openstack/openstack-ansible-os_aodh refs/changes/03/664303/1"}}},"commit":{"parents":[{"commit":"ac4fe8962ba21fb1e4d9173ee357204562ec343c","subject":"OpenDev Migration Patch","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/openstack-ansible-os_aodh/commit/ac4fe8962ba21fb1e4d9173ee357204562ec343c"}]}],"author":{"name":"OpenStack Proposal Bot","email":"openstack-infra@lists.openstack.org","date":"2019-06-10 15:43:50.000000000","tz":0},"committer":{"name":"OpenStack Proposal Bot","email":"openstack-infra@lists.openstack.org","date":"2019-06-10 15:43:50.000000000","tz":0},"subject":"Updated from OpenStack Ansible Tests","message":"Updated from OpenStack Ansible Tests\n\nChange-Id: I5c828c26c4cf77f66aa9dd3dcd257263b91af090\n","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/openstack-ansible-os_aodh/commit/52a70e898b71247d59623d4559b3bda748b22be6"}],"resolve_conflicts_web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/openstack-ansible-os_aodh/commit/52a70e898b71247d59623d4559b3bda748b22be6"}]},"branch":"refs/heads/stable/rocky"}},"requirements":[],"submit_records":[],"submit_requirements":[]}
