)]}'
{"id":"openstack%2Fopenstack-ansible-os_aodh~664410","triplet_id":"openstack%2Fopenstack-ansible-os_aodh~master~I708dfd69dbb68240fc1205893fe8c4870476da60","project":"openstack/openstack-ansible-os_aodh","branch":"master","topic":"ubuntu-distro-install","hashtags":[],"change_id":"I708dfd69dbb68240fc1205893fe8c4870476da60","subject":"Fix distro installs on Ubuntu","status":"MERGED","created":"2019-06-10 17:54:53.000000000","updated":"2019-06-11 22:38:25.000000000","submitted":"2019-06-11 22:31:16.000000000","submitter":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"total_comment_count":0,"unresolved_comment_count":0,"has_review_started":true,"submission_id":"664410-1560292276396-d9eb55b5","meta_rev_id":"a8a885acc24e4369ca2adffa1fc07a1353ebf1dd","_number":664410,"virtual_id_number":664410,"owner":{"_account_id":14805,"name":"Jimmy McCrory","email":"jimmy.mccrory@gmail.com","username":"jimmy-mccrory"},"actions":{},"labels":{"Verified":{"approved":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"all":[{"value":0,"_account_id":17799,"name":"Logan V","email":"logan2211@gmail.com","username":"Logan2211"},{"value":0,"_account_id":17130,"name":"melissaml","email":"ma.lei@99cloud.net","username":"malei"},{"value":2,"date":"2019-06-11 22:31:16.000000000","permitted_voting_range":{"min":2,"max":2},"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},{"value":0,"_account_id":25023,"name":"Jonathan Rosser","email":"jonathan.rosser@rd.bbc.co.uk","username":"jrosser"}],"values":{"-2":"Fails","-1":"Doesn\u0027t seem to work"," 0":"No score","+1":"Works for me","+2":"Verified"},"description":"","default_value":0,"optional":true},"Code-Review":{"approved":{"_account_id":17799,"name":"Logan V","email":"logan2211@gmail.com","username":"Logan2211"},"all":[{"value":2,"date":"2019-06-11 20:18:38.000000000","permitted_voting_range":{"min":2,"max":2},"_account_id":17799,"name":"Logan V","email":"logan2211@gmail.com","username":"Logan2211"},{"value":1,"date":"2019-06-11 05:48:41.000000000","permitted_voting_range":{"min":1,"max":1},"_account_id":17130,"name":"melissaml","email":"ma.lei@99cloud.net","username":"malei"},{"value":0,"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},{"value":2,"date":"2019-06-10 21:00:45.000000000","permitted_voting_range":{"min":2,"max":2},"_account_id":25023,"name":"Jonathan Rosser","email":"jonathan.rosser@rd.bbc.co.uk","username":"jrosser"}],"values":{"-2":"Do not merge","-1":"This patch needs further work before it can be merged"," 0":"No score","+1":"Looks good to me, but someone else must approve","+2":"Looks good to me (core reviewer)"},"description":"","default_value":0,"optional":true},"Workflow":{"approved":{"_account_id":17799,"name":"Logan V","email":"logan2211@gmail.com","username":"Logan2211"},"all":[{"value":1,"date":"2019-06-11 20:18:38.000000000","permitted_voting_range":{"min":1,"max":1},"_account_id":17799,"name":"Logan V","email":"logan2211@gmail.com","username":"Logan2211"},{"value":0,"_account_id":17130,"name":"melissaml","email":"ma.lei@99cloud.net","username":"malei"},{"value":0,"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},{"value":0,"_account_id":25023,"name":"Jonathan Rosser","email":"jonathan.rosser@rd.bbc.co.uk","username":"jrosser"}],"values":{"-1":"Work in progress"," 0":"Ready for reviews","+1":"Approved"},"description":"","default_value":0,"optional":true},"Backport-Candidate":{"all":[{"value":0,"_account_id":17799,"name":"Logan V","email":"logan2211@gmail.com","username":"Logan2211"},{"value":0,"_account_id":17130,"name":"melissaml","email":"ma.lei@99cloud.net","username":"malei"},{"value":0,"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},{"value":0,"_account_id":25023,"name":"Jonathan Rosser","email":"jonathan.rosser@rd.bbc.co.uk","username":"jrosser"}],"values":{"-1":"Not A Backport Candidate"," 0":"Backport Review Needed","+1":"Proposed Backport"},"description":"","default_value":0,"optional":true}},"removable_reviewers":[],"reviewers":{"REVIEWER":[{"_account_id":17130,"name":"melissaml","email":"ma.lei@99cloud.net","username":"malei"},{"_account_id":17799,"name":"Logan V","email":"logan2211@gmail.com","username":"Logan2211"},{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},{"_account_id":25023,"name":"Jonathan Rosser","email":"jonathan.rosser@rd.bbc.co.uk","username":"jrosser"}]},"pending_reviewers":{},"reviewer_updates":[{"updated":"2019-06-10 21:00:45.000000000","updated_by":{"_account_id":25023,"name":"Jonathan Rosser","email":"jonathan.rosser@rd.bbc.co.uk","username":"jrosser"},"reviewer":{"_account_id":25023,"name":"Jonathan Rosser","email":"jonathan.rosser@rd.bbc.co.uk","username":"jrosser"},"state":"REVIEWER"},{"updated":"2019-06-11 05:48:41.000000000","updated_by":{"_account_id":17130,"name":"melissaml","email":"ma.lei@99cloud.net","username":"malei"},"reviewer":{"_account_id":17130,"name":"melissaml","email":"ma.lei@99cloud.net","username":"malei"},"state":"REVIEWER"},{"updated":"2019-06-11 20:18:38.000000000","updated_by":{"_account_id":17799,"name":"Logan V","email":"logan2211@gmail.com","username":"Logan2211"},"reviewer":{"_account_id":17799,"name":"Logan V","email":"logan2211@gmail.com","username":"Logan2211"},"state":"REVIEWER"},{"updated":"2019-06-11 22:31:16.000000000","updated_by":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"reviewer":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"state":"REVIEWER"}],"messages":[{"id":"6f87fc48d6aa75345381060d8ee8f8040e48f6c1","author":{"_account_id":14805,"name":"Jimmy McCrory","email":"jimmy.mccrory@gmail.com","username":"jimmy-mccrory"},"date":"2019-06-10 17:54:53.000000000","message":"Uploaded patch set 1.","accounts_in_message":[],"_revision_number":1},{"id":"695b772b090ae788606e96b10a6d82d3c0ae6ae5","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2019-06-10 19:48:55.000000000","message":"Patch Set 1: Verified+1\n\nBuild succeeded (check pipeline).\n\n- openstack-ansible-linters http://logs.openstack.org/10/664410/1/check/openstack-ansible-linters/594950e/ : SUCCESS in 6m 22s\n- openstack-ansible-deploy-aio_metal-debian-stable http://logs.openstack.org/10/664410/1/check/openstack-ansible-deploy-aio_metal-debian-stable/23bac19/ : SUCCESS in 1h 25m 32s\n- openstack-ansible-deploy-aio_metal-centos-7 http://logs.openstack.org/10/664410/1/check/openstack-ansible-deploy-aio_metal-centos-7/44a43c0/ : SUCCESS in 1h 47m 45s\n- openstack-ansible-deploy-aio_metal-opensuse-150 http://logs.openstack.org/10/664410/1/check/openstack-ansible-deploy-aio_metal-opensuse-150/66c70c4/ : SUCCESS in 1h 33m 09s\n- openstack-ansible-deploy-aio_metal-ubuntu-bionic http://logs.openstack.org/10/664410/1/check/openstack-ansible-deploy-aio_metal-ubuntu-bionic/dd602be/ : SUCCESS in 1h 19m 05s\n- openstack-ansible-deploy-aio_distro_metal-centos-7 http://logs.openstack.org/10/664410/1/check/openstack-ansible-deploy-aio_distro_metal-centos-7/3af9ca3/ : SUCCESS in 1h 14m 24s\n- openstack-ansible-deploy-aio_telemetry_metal-centos-7 http://logs.openstack.org/10/664410/1/check/openstack-ansible-deploy-aio_telemetry_metal-centos-7/0fd4634/ : SUCCESS in 1h 40m 05s\n- openstack-ansible-deploy-aio_telemetry_metal-debian-stable http://logs.openstack.org/10/664410/1/check/openstack-ansible-deploy-aio_telemetry_metal-debian-stable/fb57e9d/ : SUCCESS in 1h 15m 56s\n- openstack-ansible-deploy-aio_telemetry_metal-ubuntu-bionic http://logs.openstack.org/10/664410/1/check/openstack-ansible-deploy-aio_telemetry_metal-ubuntu-bionic/1a6e639/ : SUCCESS in 1h 14m 11s\n- openstack-tox-docs http://logs.openstack.org/10/664410/1/check/openstack-tox-docs/00049b8/html/ : SUCCESS in 3m 31s","accounts_in_message":[],"_revision_number":1},{"id":"ed3c9664bdecf08efb07001bf30c090534248489","author":{"_account_id":25023,"name":"Jonathan Rosser","email":"jonathan.rosser@rd.bbc.co.uk","username":"jrosser"},"date":"2019-06-10 21:00:45.000000000","message":"Patch Set 1: Code-Review+2","accounts_in_message":[],"_revision_number":1},{"id":"409ea0c36361cb8b904fcebe0d1035298aed8771","author":{"_account_id":17130,"name":"melissaml","email":"ma.lei@99cloud.net","username":"malei"},"date":"2019-06-11 05:48:41.000000000","message":"Patch Set 1: Code-Review+1","accounts_in_message":[],"_revision_number":1},{"id":"4d1b4907853d413527c6ff4b42ffd472c2eec08c","author":{"_account_id":17799,"name":"Logan V","email":"logan2211@gmail.com","username":"Logan2211"},"date":"2019-06-11 20:18:38.000000000","message":"Patch Set 1: Code-Review+2 Workflow+1","accounts_in_message":[],"_revision_number":1},{"id":"0f1ef3bf01b1f2d1069c947ade395825680e9dbe","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2019-06-11 20:18:50.000000000","message":"Patch Set 1: -Verified\n\nStarting gate jobs.","accounts_in_message":[],"_revision_number":1},{"id":"90dce3d04287a396cd1bb45cc29cba910935e47d","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2019-06-11 22:31:16.000000000","message":"Change has been successfully merged by Zuul","accounts_in_message":[],"_revision_number":1},{"id":"2b43f8e350aa0b3e0c3b88b0212cc65bb8a34571","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2019-06-11 22:31:16.000000000","message":"Patch Set 1: Verified+2\n\nBuild succeeded (gate pipeline).\n\n- openstack-ansible-linters http://logs.openstack.org/10/664410/1/gate/openstack-ansible-linters/c161ef4/ : SUCCESS in 4m 21s\n- openstack-ansible-deploy-aio_metal-debian-stable http://logs.openstack.org/10/664410/1/gate/openstack-ansible-deploy-aio_metal-debian-stable/420e766/ : SUCCESS in 1h 11m 13s\n- openstack-ansible-deploy-aio_metal-centos-7 http://logs.openstack.org/10/664410/1/gate/openstack-ansible-deploy-aio_metal-centos-7/ba5caed/ : SUCCESS in 1h 34m 35s\n- openstack-ansible-deploy-aio_metal-opensuse-150 http://logs.openstack.org/10/664410/1/gate/openstack-ansible-deploy-aio_metal-opensuse-150/460a047/ : SUCCESS in 1h 33m 04s\n- openstack-ansible-deploy-aio_metal-ubuntu-bionic http://logs.openstack.org/10/664410/1/gate/openstack-ansible-deploy-aio_metal-ubuntu-bionic/30c651d/ : SUCCESS in 1h 12m 32s\n- openstack-ansible-deploy-aio_distro_metal-centos-7 http://logs.openstack.org/10/664410/1/gate/openstack-ansible-deploy-aio_distro_metal-centos-7/747b4df/ : SUCCESS in 1h 12m 17s\n- openstack-ansible-deploy-aio_telemetry_metal-centos-7 http://logs.openstack.org/10/664410/1/gate/openstack-ansible-deploy-aio_telemetry_metal-centos-7/40c5236/ : SUCCESS in 2h 04m 32s\n- openstack-ansible-deploy-aio_telemetry_metal-debian-stable http://logs.openstack.org/10/664410/1/gate/openstack-ansible-deploy-aio_telemetry_metal-debian-stable/e19b7b7/ : SUCCESS in 1h 24m 11s\n- openstack-ansible-deploy-aio_telemetry_metal-ubuntu-bionic http://logs.openstack.org/10/664410/1/gate/openstack-ansible-deploy-aio_telemetry_metal-ubuntu-bionic/76561e6/ : SUCCESS in 1h 09m 29s\n- openstack-tox-docs http://logs.openstack.org/10/664410/1/gate/openstack-tox-docs/6615ca4/html/ : SUCCESS in 2m 50s","accounts_in_message":[],"_revision_number":1},{"id":"6153cd512258c62e7476efcf0d6d8cba83b4b4b2","author":{"_account_id":14805,"name":"Jimmy McCrory","email":"jimmy.mccrory@gmail.com","username":"jimmy-mccrory"},"date":"2019-06-11 22:38:25.000000000","message":"Patch Set 1: Cherry Picked\n\nThis patchset was cherry picked to branch stable/stein as commit 7b3e2b09a31f2a7352541c9eb17635aff80102b4","accounts_in_message":[],"_revision_number":1}],"current_revision_number":1,"current_revision":"297b4e4ecd736a68b4e4b9afc817592ea97850b5","revisions":{"297b4e4ecd736a68b4e4b9afc817592ea97850b5":{"kind":"REWORK","_number":1,"created":"2019-06-10 17:54:53.000000000","uploader":{"_account_id":14805,"name":"Jimmy McCrory","email":"jimmy.mccrory@gmail.com","username":"jimmy-mccrory"},"ref":"refs/changes/10/664410/1","fetch":{"anonymous http":{"url":"https://review.opendev.org/openstack/openstack-ansible-os_aodh","ref":"refs/changes/10/664410/1","commands":{"Checkout":"git fetch https://review.opendev.org/openstack/openstack-ansible-os_aodh refs/changes/10/664410/1 \u0026\u0026 git checkout FETCH_HEAD","Cherry Pick":"git fetch https://review.opendev.org/openstack/openstack-ansible-os_aodh refs/changes/10/664410/1 \u0026\u0026 git cherry-pick FETCH_HEAD","Format Patch":"git fetch https://review.opendev.org/openstack/openstack-ansible-os_aodh refs/changes/10/664410/1 \u0026\u0026 git format-patch -1 --stdout FETCH_HEAD","Pull":"git pull https://review.opendev.org/openstack/openstack-ansible-os_aodh refs/changes/10/664410/1"}}},"commit":{"parents":[{"commit":"7224c37af7bd4658edb93cb91d5f498c6b0a6b7c","subject":"db_setup: refactor database setup to a common file","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/openstack-ansible-os_aodh/commit/7224c37af7bd4658edb93cb91d5f498c6b0a6b7c"}]}],"author":{"name":"Jimmy McCrory","email":"jimmy.mccrory@gmail.com","date":"2019-06-10 17:53:33.000000000","tz":-420},"committer":{"name":"Jimmy McCrory","email":"jimmy.mccrory@gmail.com","date":"2019-06-10 17:53:33.000000000","tz":-420},"subject":"Fix distro installs on Ubuntu","message":"Fix distro installs on Ubuntu\n\nBeginning in the Stein release, Ubuntu distro packages are now using\nPython3.\n\nInstall the \u0027python3-aodh\u0027 package instead of individual service\npackages since they provide service config files that conflict\nwith the ones OSA provides.\n\nChange-Id: I708dfd69dbb68240fc1205893fe8c4870476da60\n","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/openstack-ansible-os_aodh/commit/297b4e4ecd736a68b4e4b9afc817592ea97850b5"}],"resolve_conflicts_web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/openstack-ansible-os_aodh/commit/297b4e4ecd736a68b4e4b9afc817592ea97850b5"}]},"branch":"refs/heads/master"}},"requirements":[],"submit_records":[],"submit_requirements":[]}
