)]}'
{"id":"openstack%2Fopenstack-ansible-os_aodh~664713","triplet_id":"openstack%2Fopenstack-ansible-os_aodh~master~I55dd2ba009b8962ca0ea3e96fc4ea5a35f707286","project":"openstack/openstack-ansible-os_aodh","branch":"master","topic":"openstack/openstack-ansible-tests/sync-tests","hashtags":[],"change_id":"I55dd2ba009b8962ca0ea3e96fc4ea5a35f707286","subject":"Updated from OpenStack Ansible Tests","status":"MERGED","created":"2019-06-11 22:16:59.000000000","updated":"2019-06-19 21:09:15.000000000","submitted":"2019-06-19 21:09:15.000000000","submitter":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"total_comment_count":0,"unresolved_comment_count":0,"has_review_started":true,"submission_id":"664713-1560978555736-96f6c04a","meta_rev_id":"e74f2425cd39b7349e08db1f221d06f548adf2eb","_number":664713,"virtual_id_number":664713,"owner":{"_account_id":11131,"name":"OpenStack Proposal Bot","username":"proposal-bot","tags":["SERVICE_USER"]},"actions":{},"labels":{"Verified":{"approved":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"all":[{"value":0,"_account_id":17799,"name":"Logan V","email":"logan2211@gmail.com","username":"Logan2211"},{"value":0,"_account_id":1004,"name":"Mohammed Naser","email":"mnaser@vexxhost.com","username":"mnaser"},{"value":2,"date":"2019-06-19 21:09:15.000000000","permitted_voting_range":{"min":2,"max":2},"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]}],"values":{"-2":"Fails","-1":"Doesn\u0027t seem to work"," 0":"No score","+1":"Works for me","+2":"Verified"},"description":"","default_value":0,"optional":true},"Code-Review":{"approved":{"_account_id":17799,"name":"Logan V","email":"logan2211@gmail.com","username":"Logan2211"},"all":[{"value":2,"date":"2019-06-18 18:54:53.000000000","permitted_voting_range":{"min":2,"max":2},"_account_id":17799,"name":"Logan V","email":"logan2211@gmail.com","username":"Logan2211"},{"value":2,"date":"2019-06-19 19:18:21.000000000","permitted_voting_range":{"min":2,"max":2},"_account_id":1004,"name":"Mohammed Naser","email":"mnaser@vexxhost.com","username":"mnaser"},{"value":0,"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]}],"values":{"-2":"Do not merge","-1":"This patch needs further work before it can be merged"," 0":"No score","+1":"Looks good to me, but someone else must approve","+2":"Looks good to me (core reviewer)"},"description":"","default_value":0,"optional":true},"Workflow":{"approved":{"_account_id":1004,"name":"Mohammed Naser","email":"mnaser@vexxhost.com","username":"mnaser"},"all":[{"value":0,"_account_id":17799,"name":"Logan V","email":"logan2211@gmail.com","username":"Logan2211"},{"value":1,"date":"2019-06-19 19:18:21.000000000","permitted_voting_range":{"min":1,"max":1},"_account_id":1004,"name":"Mohammed Naser","email":"mnaser@vexxhost.com","username":"mnaser"},{"value":0,"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]}],"values":{"-1":"Work in progress"," 0":"Ready for reviews","+1":"Approved"},"description":"","default_value":0,"optional":true},"Backport-Candidate":{"all":[{"value":0,"_account_id":17799,"name":"Logan V","email":"logan2211@gmail.com","username":"Logan2211"},{"value":0,"_account_id":1004,"name":"Mohammed Naser","email":"mnaser@vexxhost.com","username":"mnaser"},{"value":0,"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]}],"values":{"-1":"Not A Backport Candidate"," 0":"Backport Review Needed","+1":"Proposed Backport"},"description":"","default_value":0,"optional":true}},"removable_reviewers":[],"reviewers":{"REVIEWER":[{"_account_id":1004,"name":"Mohammed Naser","email":"mnaser@vexxhost.com","username":"mnaser"},{"_account_id":17799,"name":"Logan V","email":"logan2211@gmail.com","username":"Logan2211"},{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]}]},"pending_reviewers":{},"reviewer_updates":[{"updated":"2019-06-18 18:54:53.000000000","updated_by":{"_account_id":17799,"name":"Logan V","email":"logan2211@gmail.com","username":"Logan2211"},"reviewer":{"_account_id":17799,"name":"Logan V","email":"logan2211@gmail.com","username":"Logan2211"},"state":"REVIEWER"},{"updated":"2019-06-19 19:18:21.000000000","updated_by":{"_account_id":1004,"name":"Mohammed Naser","email":"mnaser@vexxhost.com","username":"mnaser"},"reviewer":{"_account_id":1004,"name":"Mohammed Naser","email":"mnaser@vexxhost.com","username":"mnaser"},"state":"REVIEWER"},{"updated":"2019-06-19 21:09:15.000000000","updated_by":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"reviewer":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"state":"REVIEWER"}],"messages":[{"id":"827e064cfdffc01820e2b897c68896da55f1a9e0","author":{"_account_id":11131,"name":"OpenStack Proposal Bot","username":"proposal-bot","tags":["SERVICE_USER"]},"date":"2019-06-11 22:16:59.000000000","message":"Uploaded patch set 1.","accounts_in_message":[],"_revision_number":1},{"id":"fa31035ebeffbd9de087184987b282266198e352","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2019-06-12 00:04:57.000000000","message":"Patch Set 1: Verified-1\n\nBuild failed (check pipeline).  For information on how to proceed, see\nhttp://docs.openstack.org/infra/manual/developers.html#automated-testing\n\n\n- openstack-ansible-linters http://logs.openstack.org/13/664713/1/check/openstack-ansible-linters/eef7927/ : SUCCESS in 5m 41s\n- openstack-ansible-deploy-aio_metal-debian-stable http://logs.openstack.org/13/664713/1/check/openstack-ansible-deploy-aio_metal-debian-stable/b0215d9/ : SUCCESS in 1h 21m 35s\n- openstack-ansible-deploy-aio_metal-centos-7 http://logs.openstack.org/13/664713/1/check/openstack-ansible-deploy-aio_metal-centos-7/474e908/ : FAILURE in 15m 01s\n- openstack-ansible-deploy-aio_metal-opensuse-150 http://logs.openstack.org/13/664713/1/check/openstack-ansible-deploy-aio_metal-opensuse-150/3dc40b6/ : SUCCESS in 1h 15m 03s\n- openstack-ansible-deploy-aio_metal-ubuntu-bionic http://logs.openstack.org/13/664713/1/check/openstack-ansible-deploy-aio_metal-ubuntu-bionic/afa9e80/ : SUCCESS in 1h 05m 17s\n- openstack-ansible-deploy-aio_distro_metal-centos-7 http://logs.openstack.org/13/664713/1/check/openstack-ansible-deploy-aio_distro_metal-centos-7/565a4fa/ : SUCCESS in 1h 13m 07s\n- openstack-ansible-deploy-aio_telemetry_metal-centos-7 http://logs.openstack.org/13/664713/1/check/openstack-ansible-deploy-aio_telemetry_metal-centos-7/05c4379/ : SUCCESS in 1h 41m 18s\n- openstack-ansible-deploy-aio_telemetry_metal-debian-stable http://logs.openstack.org/13/664713/1/check/openstack-ansible-deploy-aio_telemetry_metal-debian-stable/91947b4/ : SUCCESS in 1h 04m 11s\n- openstack-ansible-deploy-aio_telemetry_metal-ubuntu-bionic http://logs.openstack.org/13/664713/1/check/openstack-ansible-deploy-aio_telemetry_metal-ubuntu-bionic/4eb999f/ : SUCCESS in 1h 36m 43s\n- openstack-tox-docs http://logs.openstack.org/13/664713/1/check/openstack-tox-docs/6a84461/html/ : SUCCESS in 2m 45s","accounts_in_message":[],"_revision_number":1},{"id":"e621414f728ff690271cf357b35d8c121993eda9","author":{"_account_id":11131,"name":"OpenStack Proposal Bot","username":"proposal-bot","tags":["SERVICE_USER"]},"date":"2019-06-18 18:15:20.000000000","message":"Uploaded patch set 2.","accounts_in_message":[],"_revision_number":2},{"id":"d37d47671f85a87f319bd7349155e1b8d1ff8690","author":{"_account_id":17799,"name":"Logan V","email":"logan2211@gmail.com","username":"Logan2211"},"date":"2019-06-18 18:54:53.000000000","message":"Patch Set 2: Code-Review+2","accounts_in_message":[],"_revision_number":2},{"id":"97e7b1fd5692673eb9674d48e6b4796e141f1f28","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2019-06-18 20:13:25.000000000","message":"Patch Set 2: Verified+1\n\nBuild succeeded (check pipeline).\n\n- openstack-ansible-linters http://logs.openstack.org/13/664713/2/check/openstack-ansible-linters/ce1a365/ : SUCCESS in 5m 55s\n- openstack-ansible-deploy-aio_metal-debian-stable http://logs.openstack.org/13/664713/2/check/openstack-ansible-deploy-aio_metal-debian-stable/21465fe/ : SUCCESS in 1h 27m 20s\n- openstack-ansible-deploy-aio_metal-centos-7 http://logs.openstack.org/13/664713/2/check/openstack-ansible-deploy-aio_metal-centos-7/2db005a/ : SUCCESS in 1h 33m 57s\n- openstack-ansible-deploy-aio_metal-opensuse-150 http://logs.openstack.org/13/664713/2/check/openstack-ansible-deploy-aio_metal-opensuse-150/16f5985/ : SUCCESS in 1h 27m 02s\n- openstack-ansible-deploy-aio_metal-ubuntu-bionic http://logs.openstack.org/13/664713/2/check/openstack-ansible-deploy-aio_metal-ubuntu-bionic/4498ae6/ : SUCCESS in 1h 12m 04s\n- openstack-ansible-deploy-aio_distro_metal-centos-7 http://logs.openstack.org/13/664713/2/check/openstack-ansible-deploy-aio_distro_metal-centos-7/e2ee87a/ : SUCCESS in 1h 09m 36s\n- openstack-ansible-deploy-aio_telemetry_metal-centos-7 http://logs.openstack.org/13/664713/2/check/openstack-ansible-deploy-aio_telemetry_metal-centos-7/736dfda/ : SUCCESS in 1h 38m 50s\n- openstack-ansible-deploy-aio_telemetry_metal-debian-stable http://logs.openstack.org/13/664713/2/check/openstack-ansible-deploy-aio_telemetry_metal-debian-stable/5db40f1/ : SUCCESS in 1h 20m 47s\n- openstack-ansible-deploy-aio_telemetry_metal-ubuntu-bionic http://logs.openstack.org/13/664713/2/check/openstack-ansible-deploy-aio_telemetry_metal-ubuntu-bionic/7083949/ : SUCCESS in 1h 11m 31s\n- openstack-tox-docs http://logs.openstack.org/13/664713/2/check/openstack-tox-docs/04c0585/html/ : SUCCESS in 2m 54s","accounts_in_message":[],"_revision_number":2},{"id":"13ad7fb5976d8e84d8d098a3d7430fea0400f02f","author":{"_account_id":1004,"name":"Mohammed Naser","email":"mnaser@vexxhost.com","username":"mnaser"},"date":"2019-06-19 19:18:21.000000000","message":"Patch Set 2: Code-Review+2 Workflow+1","accounts_in_message":[],"_revision_number":2},{"id":"39389c5a34016a0d377275832e5a9f7bf6dc29cb","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2019-06-19 19:18:33.000000000","message":"Patch Set 2: -Verified\n\nStarting gate jobs.","accounts_in_message":[],"_revision_number":2},{"id":"3a2574177ff30a98cb63c3f57eaf8f0f4b48d14f","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2019-06-19 21:09:15.000000000","message":"Patch Set 2: Verified+2\n\nBuild succeeded (gate pipeline).\n\n- openstack-ansible-linters http://logs.openstack.org/13/664713/2/gate/openstack-ansible-linters/ca52b9e/ : SUCCESS in 5m 04s\n- openstack-ansible-deploy-aio_metal-debian-stable http://logs.openstack.org/13/664713/2/gate/openstack-ansible-deploy-aio_metal-debian-stable/fffd6a3/ : SUCCESS in 1h 16m 14s\n- openstack-ansible-deploy-aio_metal-centos-7 http://logs.openstack.org/13/664713/2/gate/openstack-ansible-deploy-aio_metal-centos-7/c32bf27/ : SUCCESS in 1h 38m 57s\n- openstack-ansible-deploy-aio_metal-opensuse-150 http://logs.openstack.org/13/664713/2/gate/openstack-ansible-deploy-aio_metal-opensuse-150/d804486/ : SUCCESS in 1h 29m 37s\n- openstack-ansible-deploy-aio_metal-ubuntu-bionic http://logs.openstack.org/13/664713/2/gate/openstack-ansible-deploy-aio_metal-ubuntu-bionic/3fc1fcc/ : SUCCESS in 1h 04m 33s\n- openstack-ansible-deploy-aio_distro_metal-centos-7 http://logs.openstack.org/13/664713/2/gate/openstack-ansible-deploy-aio_distro_metal-centos-7/fdd98a2/ : SUCCESS in 1h 30m 11s\n- openstack-ansible-deploy-aio_telemetry_metal-centos-7 http://logs.openstack.org/13/664713/2/gate/openstack-ansible-deploy-aio_telemetry_metal-centos-7/fc71cb8/ : SUCCESS in 1h 36m 53s\n- openstack-ansible-deploy-aio_telemetry_metal-debian-stable http://logs.openstack.org/13/664713/2/gate/openstack-ansible-deploy-aio_telemetry_metal-debian-stable/64e9c32/ : SUCCESS in 1h 28m 16s\n- openstack-ansible-deploy-aio_telemetry_metal-ubuntu-bionic http://logs.openstack.org/13/664713/2/gate/openstack-ansible-deploy-aio_telemetry_metal-ubuntu-bionic/8735798/ : SUCCESS in 1h 08m 51s\n- openstack-tox-docs http://logs.openstack.org/13/664713/2/gate/openstack-tox-docs/92cefb9/html/ : SUCCESS in 2m 57s","accounts_in_message":[],"_revision_number":2},{"id":"138b91bd9016171f790cb7517868f772264027ea","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2019-06-19 21:09:15.000000000","message":"Change has been successfully merged by Zuul","accounts_in_message":[],"_revision_number":2}],"current_revision_number":2,"current_revision":"14891358c8c650d6f47ee374227a07339b087977","revisions":{"d3e42835f47e37b324c54725ded3154d782e0543":{"kind":"REWORK","_number":1,"created":"2019-06-11 22:16:59.000000000","uploader":{"_account_id":11131,"name":"OpenStack Proposal Bot","username":"proposal-bot","tags":["SERVICE_USER"]},"ref":"refs/changes/13/664713/1","fetch":{"anonymous http":{"url":"https://review.opendev.org/openstack/openstack-ansible-os_aodh","ref":"refs/changes/13/664713/1","commands":{"Checkout":"git fetch https://review.opendev.org/openstack/openstack-ansible-os_aodh refs/changes/13/664713/1 \u0026\u0026 git checkout FETCH_HEAD","Cherry Pick":"git fetch https://review.opendev.org/openstack/openstack-ansible-os_aodh refs/changes/13/664713/1 \u0026\u0026 git cherry-pick FETCH_HEAD","Format Patch":"git fetch https://review.opendev.org/openstack/openstack-ansible-os_aodh refs/changes/13/664713/1 \u0026\u0026 git format-patch -1 --stdout FETCH_HEAD","Pull":"git pull https://review.opendev.org/openstack/openstack-ansible-os_aodh refs/changes/13/664713/1"}}},"commit":{"parents":[{"commit":"7224c37af7bd4658edb93cb91d5f498c6b0a6b7c","subject":"db_setup: refactor database setup to a common file","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/openstack-ansible-os_aodh/commit/7224c37af7bd4658edb93cb91d5f498c6b0a6b7c"}]}],"author":{"name":"OpenStack Proposal Bot","email":"openstack-infra@lists.openstack.org","date":"2019-06-11 22:16:57.000000000","tz":0},"committer":{"name":"OpenStack Proposal Bot","email":"openstack-infra@lists.openstack.org","date":"2019-06-11 22:16:57.000000000","tz":0},"subject":"Updated from OpenStack Ansible Tests","message":"Updated from OpenStack Ansible Tests\n\nChange-Id: I55dd2ba009b8962ca0ea3e96fc4ea5a35f707286\n","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/openstack-ansible-os_aodh/commit/d3e42835f47e37b324c54725ded3154d782e0543"}],"resolve_conflicts_web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/openstack-ansible-os_aodh/commit/d3e42835f47e37b324c54725ded3154d782e0543"}]},"branch":"refs/heads/master"},"14891358c8c650d6f47ee374227a07339b087977":{"kind":"REWORK","_number":2,"created":"2019-06-18 18:15:20.000000000","uploader":{"_account_id":11131,"name":"OpenStack Proposal Bot","username":"proposal-bot","tags":["SERVICE_USER"]},"ref":"refs/changes/13/664713/2","fetch":{"anonymous http":{"url":"https://review.opendev.org/openstack/openstack-ansible-os_aodh","ref":"refs/changes/13/664713/2","commands":{"Checkout":"git fetch https://review.opendev.org/openstack/openstack-ansible-os_aodh refs/changes/13/664713/2 \u0026\u0026 git checkout FETCH_HEAD","Cherry Pick":"git fetch https://review.opendev.org/openstack/openstack-ansible-os_aodh refs/changes/13/664713/2 \u0026\u0026 git cherry-pick FETCH_HEAD","Format Patch":"git fetch https://review.opendev.org/openstack/openstack-ansible-os_aodh refs/changes/13/664713/2 \u0026\u0026 git format-patch -1 --stdout FETCH_HEAD","Pull":"git pull https://review.opendev.org/openstack/openstack-ansible-os_aodh refs/changes/13/664713/2"}}},"commit":{"parents":[{"commit":"297b4e4ecd736a68b4e4b9afc817592ea97850b5","subject":"Fix distro installs on Ubuntu","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/openstack-ansible-os_aodh/commit/297b4e4ecd736a68b4e4b9afc817592ea97850b5"}]}],"author":{"name":"OpenStack Proposal Bot","email":"openstack-infra@lists.openstack.org","date":"2019-06-18 18:15:19.000000000","tz":0},"committer":{"name":"OpenStack Proposal Bot","email":"openstack-infra@lists.openstack.org","date":"2019-06-18 18:15:19.000000000","tz":0},"subject":"Updated from OpenStack Ansible Tests","message":"Updated from OpenStack Ansible Tests\n\nChange-Id: I55dd2ba009b8962ca0ea3e96fc4ea5a35f707286\n","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/openstack-ansible-os_aodh/commit/14891358c8c650d6f47ee374227a07339b087977"}],"resolve_conflicts_web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/openstack-ansible-os_aodh/commit/14891358c8c650d6f47ee374227a07339b087977"}]},"branch":"refs/heads/master"}},"requirements":[],"submit_records":[],"submit_requirements":[]}
