)]}'
{"id":"openstack%2Fopenstack-ansible-os_aodh~671590","triplet_id":"openstack%2Fopenstack-ansible-os_aodh~master~I407267fc32a4a755ab2ee91095ddc401f68a1b19","project":"openstack/openstack-ansible-os_aodh","branch":"master","topic":"openstack/openstack-ansible-tests/sync-tests","hashtags":[],"change_id":"I407267fc32a4a755ab2ee91095ddc401f68a1b19","subject":"Updated from OpenStack Ansible Tests","status":"MERGED","created":"2019-07-18 22:17:32.000000000","updated":"2019-08-23 10:55:10.000000000","submitted":"2019-08-23 10:53:30.000000000","submitter":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"total_comment_count":0,"unresolved_comment_count":0,"has_review_started":true,"submission_id":"671590-1566557610347-d74ab91c","meta_rev_id":"089d32621f935f8ad00a3909c2a42a7519c97a48","_number":671590,"virtual_id_number":671590,"owner":{"_account_id":11131,"name":"OpenStack Proposal Bot","username":"proposal-bot","tags":["SERVICE_USER"]},"actions":{},"labels":{"Verified":{"approved":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"all":[{"value":0,"_account_id":28619,"name":"Dmitriy Rabotyagov","email":"noonedeadpunk@gmail.com","username":"noonedeadpunk"},{"value":2,"date":"2019-08-23 10:53:30.000000000","post_submit":true,"permitted_voting_range":{"min":2,"max":2},"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]}],"values":{"-2":"Fails","-1":"Doesn\u0027t seem to work"," 0":"No score","+1":"Works for me","+2":"Verified"},"description":"","default_value":0,"optional":true},"Code-Review":{"approved":{"_account_id":28619,"name":"Dmitriy Rabotyagov","email":"noonedeadpunk@gmail.com","username":"noonedeadpunk"},"all":[{"value":2,"date":"2019-08-20 10:54:17.000000000","permitted_voting_range":{"min":2,"max":2},"_account_id":28619,"name":"Dmitriy Rabotyagov","email":"noonedeadpunk@gmail.com","username":"noonedeadpunk"},{"value":0,"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]}],"values":{"-2":"Do not merge","-1":"This patch needs further work before it can be merged"," 0":"No score","+1":"Looks good to me, but someone else must approve","+2":"Looks good to me (core reviewer)"},"description":"","default_value":0,"optional":true},"Workflow":{"approved":{"_account_id":28619,"name":"Dmitriy Rabotyagov","email":"noonedeadpunk@gmail.com","username":"noonedeadpunk"},"all":[{"value":1,"date":"2019-08-20 10:54:17.000000000","permitted_voting_range":{"min":1,"max":1},"_account_id":28619,"name":"Dmitriy Rabotyagov","email":"noonedeadpunk@gmail.com","username":"noonedeadpunk"},{"value":0,"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]}],"values":{"-1":"Work in progress"," 0":"Ready for reviews","+1":"Approved"},"description":"","default_value":0,"optional":true},"Backport-Candidate":{"all":[{"value":0,"_account_id":28619,"name":"Dmitriy Rabotyagov","email":"noonedeadpunk@gmail.com","username":"noonedeadpunk"},{"value":0,"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]}],"values":{"-1":"Not A Backport Candidate"," 0":"Backport Review Needed","+1":"Proposed Backport"},"description":"","default_value":0,"optional":true}},"removable_reviewers":[],"reviewers":{"REVIEWER":[{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},{"_account_id":28619,"name":"Dmitriy Rabotyagov","email":"noonedeadpunk@gmail.com","username":"noonedeadpunk"}]},"pending_reviewers":{},"reviewer_updates":[{"updated":"2019-08-20 10:54:17.000000000","updated_by":{"_account_id":28619,"name":"Dmitriy Rabotyagov","email":"noonedeadpunk@gmail.com","username":"noonedeadpunk"},"reviewer":{"_account_id":28619,"name":"Dmitriy Rabotyagov","email":"noonedeadpunk@gmail.com","username":"noonedeadpunk"},"state":"REVIEWER"},{"updated":"2019-08-23 10:53:30.000000000","updated_by":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"reviewer":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"state":"REVIEWER"}],"messages":[{"id":"2d2d4e39532270a19215b93a6866fcaa6a1b4062","author":{"_account_id":11131,"name":"OpenStack Proposal Bot","username":"proposal-bot","tags":["SERVICE_USER"]},"date":"2019-07-18 22:17:32.000000000","message":"Uploaded patch set 1.","accounts_in_message":[],"_revision_number":1},{"id":"37da4789da75bd2aa47e9795a524f2e84321edfb","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2019-07-19 00:14:27.000000000","message":"Patch Set 1: Verified-1\n\nBuild failed (check pipeline).  For information on how to proceed, see\nhttp://docs.openstack.org/infra/manual/developers.html#automated-testing\n\n\n- openstack-ansible-linters http://logs.openstack.org/90/671590/1/check/openstack-ansible-linters/51efe7f/ : SUCCESS in 6m 20s\n- openstack-ansible-deploy-aio_metal-debian-stable http://logs.openstack.org/90/671590/1/check/openstack-ansible-deploy-aio_metal-debian-stable/a5af979/ : SUCCESS in 1h 11m 59s\n- openstack-ansible-deploy-aio_metal-centos-7 http://logs.openstack.org/90/671590/1/check/openstack-ansible-deploy-aio_metal-centos-7/d8a0636/ : SUCCESS in 1h 47m 08s\n- openstack-ansible-deploy-aio_metal-opensuse-15 http://logs.openstack.org/90/671590/1/check/openstack-ansible-deploy-aio_metal-opensuse-15/7d4f417/ : SUCCESS in 1h 20m 25s\n- openstack-ansible-deploy-aio_metal-ubuntu-bionic http://logs.openstack.org/90/671590/1/check/openstack-ansible-deploy-aio_metal-ubuntu-bionic/a01897c/ : SUCCESS in 1h 16m 03s\n- openstack-ansible-deploy-aio_distro_metal-centos-7 http://logs.openstack.org/90/671590/1/check/openstack-ansible-deploy-aio_distro_metal-centos-7/bd23dc0/ : SUCCESS in 1h 18m 48s\n- openstack-ansible-deploy-aio_distro_metal-opensuse-15 http://logs.openstack.org/90/671590/1/check/openstack-ansible-deploy-aio_distro_metal-opensuse-15/4e95a90/ : FAILURE in 55m 18s (non-voting)\n- openstack-ansible-deploy-aio_distro_metal-ubuntu-bionic http://logs.openstack.org/90/671590/1/check/openstack-ansible-deploy-aio_distro_metal-ubuntu-bionic/8a33db7/ : FAILURE in 35m 37s\n- openstack-ansible-deploy-aio_telemetry_metal-centos-7 http://logs.openstack.org/90/671590/1/check/openstack-ansible-deploy-aio_telemetry_metal-centos-7/6ac4930/ : SUCCESS in 1h 48m 23s\n- openstack-ansible-deploy-aio_telemetry_metal-debian-stable http://logs.openstack.org/90/671590/1/check/openstack-ansible-deploy-aio_telemetry_metal-debian-stable/ae23b94/ : SUCCESS in 1h 06m 02s\n- openstack-ansible-deploy-aio_telemetry_metal-ubuntu-bionic http://logs.openstack.org/90/671590/1/check/openstack-ansible-deploy-aio_telemetry_metal-ubuntu-bionic/5a2c12e/ : SUCCESS in 1h 10m 04s\n- openstack-tox-docs http://logs.openstack.org/90/671590/1/check/openstack-tox-docs/3521990/html/ : SUCCESS in 4m 05s","accounts_in_message":[],"_revision_number":1},{"id":"37525d41e0df820d0ca2159218bab8e9bdd7f52b","author":{"_account_id":11131,"name":"OpenStack Proposal Bot","username":"proposal-bot","tags":["SERVICE_USER"]},"date":"2019-07-19 18:19:14.000000000","message":"Uploaded patch set 2.","accounts_in_message":[],"_revision_number":2},{"id":"c36e640e9f4fb6d61575ca6fb4403f59ffd95ff0","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2019-07-19 20:19:40.000000000","message":"Patch Set 2: Verified-1\n\nBuild failed (check pipeline).  For information on how to proceed, see\nhttp://docs.openstack.org/infra/manual/developers.html#automated-testing\n\n\n- requirements-check http://logs.openstack.org/90/671590/2/check/requirements-check/b75903c/ : SUCCESS in 2m 31s\n- openstack-ansible-linters http://logs.openstack.org/90/671590/2/check/openstack-ansible-linters/45b5b90/ : SUCCESS in 5m 51s\n- openstack-ansible-deploy-aio_metal-debian-stable http://logs.openstack.org/90/671590/2/check/openstack-ansible-deploy-aio_metal-debian-stable/6fd5388/ : SUCCESS in 1h 21m 52s\n- openstack-ansible-deploy-aio_metal-centos-7 http://logs.openstack.org/90/671590/2/check/openstack-ansible-deploy-aio_metal-centos-7/b6f9da6/ : SUCCESS in 1h 41m 11s\n- openstack-ansible-deploy-aio_metal-opensuse-15 http://logs.openstack.org/90/671590/2/check/openstack-ansible-deploy-aio_metal-opensuse-15/a6fc448/ : SUCCESS in 1h 28m 16s\n- openstack-ansible-deploy-aio_metal-ubuntu-bionic http://logs.openstack.org/90/671590/2/check/openstack-ansible-deploy-aio_metal-ubuntu-bionic/9469169/ : SUCCESS in 1h 11m 01s\n- openstack-ansible-deploy-aio_distro_metal-centos-7 http://logs.openstack.org/90/671590/2/check/openstack-ansible-deploy-aio_distro_metal-centos-7/9311564/ : SUCCESS in 1h 11m 45s\n- openstack-ansible-deploy-aio_distro_metal-opensuse-15 http://logs.openstack.org/90/671590/2/check/openstack-ansible-deploy-aio_distro_metal-opensuse-15/84bc281/ : FAILURE in 46m 20s (non-voting)\n- openstack-ansible-deploy-aio_distro_metal-ubuntu-bionic http://logs.openstack.org/90/671590/2/check/openstack-ansible-deploy-aio_distro_metal-ubuntu-bionic/1970f85/ : FAILURE in 37m 44s\n- openstack-ansible-deploy-aio_telemetry_metal-centos-7 http://logs.openstack.org/90/671590/2/check/openstack-ansible-deploy-aio_telemetry_metal-centos-7/c644f5d/ : FAILURE in 1h 49m 27s\n- openstack-ansible-deploy-aio_telemetry_metal-debian-stable http://logs.openstack.org/90/671590/2/check/openstack-ansible-deploy-aio_telemetry_metal-debian-stable/173b22d/ : SUCCESS in 1h 21m 34s\n- openstack-ansible-deploy-aio_telemetry_metal-ubuntu-bionic http://logs.openstack.org/90/671590/2/check/openstack-ansible-deploy-aio_telemetry_metal-ubuntu-bionic/0e16676/ : SUCCESS in 1h 19m 18s\n- openstack-tox-docs http://logs.openstack.org/90/671590/2/check/openstack-tox-docs/fbc6568/ : POST_FAILURE in 2m 55s\n- build-openstack-releasenotes http://logs.openstack.org/90/671590/2/check/build-openstack-releasenotes/a7264ad/html/ : SUCCESS in 3m 43s","accounts_in_message":[],"_revision_number":2},{"id":"619f10effe9afa3b316a0d46c2baa6c1d95b5663","author":{"_account_id":11131,"name":"OpenStack Proposal Bot","username":"proposal-bot","tags":["SERVICE_USER"]},"date":"2019-07-20 09:29:46.000000000","message":"Uploaded patch set 3.","accounts_in_message":[],"_revision_number":3},{"id":"b7f6050c9589f3bb0ef8814766c2635aca0bda95","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2019-07-20 11:05:32.000000000","message":"Patch Set 3: Verified-1\n\nBuild failed (check pipeline).  For information on how to proceed, see\nhttp://docs.openstack.org/infra/manual/developers.html#automated-testing\n\n\n- requirements-check http://logs.openstack.org/90/671590/3/check/requirements-check/6186aee/ : SUCCESS in 3m 04s\n- openstack-ansible-linters http://logs.openstack.org/90/671590/3/check/openstack-ansible-linters/1027a36/ : SUCCESS in 6m 33s\n- openstack-ansible-deploy-aio_metal-debian-stable http://logs.openstack.org/90/671590/3/check/openstack-ansible-deploy-aio_metal-debian-stable/320559f/ : SUCCESS in 1h 10m 34s\n- openstack-ansible-deploy-aio_metal-centos-7 http://logs.openstack.org/90/671590/3/check/openstack-ansible-deploy-aio_metal-centos-7/f0779f8/ : SUCCESS in 1h 31m 38s\n- openstack-ansible-deploy-aio_metal-opensuse-15 http://logs.openstack.org/90/671590/3/check/openstack-ansible-deploy-aio_metal-opensuse-15/b7ffb1b/ : SUCCESS in 1h 22m 56s\n- openstack-ansible-deploy-aio_metal-ubuntu-bionic http://logs.openstack.org/90/671590/3/check/openstack-ansible-deploy-aio_metal-ubuntu-bionic/2c9d39d/ : SUCCESS in 1h 11m 21s\n- openstack-ansible-deploy-aio_distro_metal-centos-7 http://logs.openstack.org/90/671590/3/check/openstack-ansible-deploy-aio_distro_metal-centos-7/d7551a3/ : SUCCESS in 1h 11m 15s\n- openstack-ansible-deploy-aio_distro_metal-opensuse-15 http://logs.openstack.org/90/671590/3/check/openstack-ansible-deploy-aio_distro_metal-opensuse-15/75e4fee/ : FAILURE in 43m 39s (non-voting)\n- openstack-ansible-deploy-aio_distro_metal-ubuntu-bionic http://logs.openstack.org/90/671590/3/check/openstack-ansible-deploy-aio_distro_metal-ubuntu-bionic/e379c0b/ : FAILURE in 31m 07s\n- openstack-ansible-deploy-aio_telemetry_metal-centos-7 http://logs.openstack.org/90/671590/3/check/openstack-ansible-deploy-aio_telemetry_metal-centos-7/57b2c48/ : SUCCESS in 1h 31m 36s\n- openstack-ansible-deploy-aio_telemetry_metal-debian-stable http://logs.openstack.org/90/671590/3/check/openstack-ansible-deploy-aio_telemetry_metal-debian-stable/e220c32/ : SUCCESS in 1h 08m 49s\n- openstack-ansible-deploy-aio_telemetry_metal-ubuntu-bionic http://logs.openstack.org/90/671590/3/check/openstack-ansible-deploy-aio_telemetry_metal-ubuntu-bionic/e77f7b4/ : SUCCESS in 1h 11m 58s\n- openstack-tox-docs http://logs.openstack.org/90/671590/3/check/openstack-tox-docs/955927a/html/ : SUCCESS in 3m 15s\n- build-openstack-releasenotes http://logs.openstack.org/90/671590/3/check/build-openstack-releasenotes/a456f06/html/ : SUCCESS in 5m 06s","accounts_in_message":[],"_revision_number":3},{"id":"4587e4f883a689126ac13035054df784463b9c7d","author":{"_account_id":11131,"name":"OpenStack Proposal Bot","username":"proposal-bot","tags":["SERVICE_USER"]},"date":"2019-07-30 19:37:12.000000000","message":"Uploaded patch set 4.","accounts_in_message":[],"_revision_number":4},{"id":"aa8ce75fb5f497c661e5d6d4fdd1c3e5b988f184","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2019-07-30 21:52:04.000000000","message":"Patch Set 4: Verified-1\n\nBuild failed (check pipeline).  For information on how to proceed, see\nhttp://docs.openstack.org/infra/manual/developers.html#automated-testing\n\n\n- requirements-check http://logs.openstack.org/90/671590/4/check/requirements-check/7acfe4a/ : SUCCESS in 3m 32s\n- openstack-ansible-linters http://logs.openstack.org/90/671590/4/check/openstack-ansible-linters/66360de/ : SUCCESS in 7m 38s\n- openstack-ansible-deploy-aio_metal-debian-stable http://logs.openstack.org/90/671590/4/check/openstack-ansible-deploy-aio_metal-debian-stable/7bf6438/ : FAILURE in 1h 26m 29s\n- openstack-ansible-deploy-aio_metal-centos-7 http://logs.openstack.org/90/671590/4/check/openstack-ansible-deploy-aio_metal-centos-7/215f233/ : SUCCESS in 1h 46m 54s\n- openstack-ansible-deploy-aio_metal-opensuse-15 http://logs.openstack.org/90/671590/4/check/openstack-ansible-deploy-aio_metal-opensuse-15/a19c0f7/ : SUCCESS in 1h 36m 40s\n- openstack-ansible-deploy-aio_metal-ubuntu-bionic http://logs.openstack.org/90/671590/4/check/openstack-ansible-deploy-aio_metal-ubuntu-bionic/2544348/ : SUCCESS in 1h 15m 02s\n- openstack-ansible-deploy-aio_distro_metal-centos-7 http://logs.openstack.org/90/671590/4/check/openstack-ansible-deploy-aio_distro_metal-centos-7/cd09e72/ : SUCCESS in 1h 12m 48s\n- openstack-ansible-deploy-aio_distro_metal-opensuse-15 http://logs.openstack.org/90/671590/4/check/openstack-ansible-deploy-aio_distro_metal-opensuse-15/eb71462/ : FAILURE in 54m 19s (non-voting)\n- openstack-ansible-deploy-aio_distro_metal-ubuntu-bionic http://logs.openstack.org/90/671590/4/check/openstack-ansible-deploy-aio_distro_metal-ubuntu-bionic/21b7c1e/ : SUCCESS in 51m 22s\n- openstack-ansible-deploy-aio_telemetry_metal-centos-7 http://logs.openstack.org/90/671590/4/check/openstack-ansible-deploy-aio_telemetry_metal-centos-7/38a00ba/ : SUCCESS in 1h 42m 58s\n- openstack-ansible-deploy-aio_telemetry_metal-debian-stable http://logs.openstack.org/90/671590/4/check/openstack-ansible-deploy-aio_telemetry_metal-debian-stable/bf96d8b/ : SUCCESS in 1h 16m 44s\n- openstack-ansible-deploy-aio_telemetry_metal-ubuntu-bionic http://logs.openstack.org/90/671590/4/check/openstack-ansible-deploy-aio_telemetry_metal-ubuntu-bionic/6c2d068/ : SUCCESS in 2h 03m 58s\n- openstack-tox-docs http://logs.openstack.org/90/671590/4/check/openstack-tox-docs/78db2ca/html/ : SUCCESS in 3m 00s\n- build-openstack-releasenotes http://logs.openstack.org/90/671590/4/check/build-openstack-releasenotes/87d97bd/html/ : SUCCESS in 3m 35s","accounts_in_message":[],"_revision_number":4},{"id":"1465e530951330806de4d830b64414744c950441","author":{"_account_id":11131,"name":"OpenStack Proposal Bot","username":"proposal-bot","tags":["SERVICE_USER"]},"date":"2019-08-20 03:05:33.000000000","message":"Uploaded patch set 5.","accounts_in_message":[],"_revision_number":5},{"id":"04cefb5c4e52478b56107224f6d21a44a2dbd193","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2019-08-20 05:11:51.000000000","message":"Patch Set 5: Verified+1\n\nBuild succeeded (check pipeline).\n\n- requirements-check https://zuul.opendev.org/t/openstack/build/00a0ef6438ca4d2dbffeb8bf31fbbe78 : SUCCESS in 3m 39s\n- openstack-ansible-linters https://zuul.opendev.org/t/openstack/build/be2e63efb682402b98e07919239081f3 : SUCCESS in 5m 58s\n- openstack-ansible-deploy-aio_metal-debian-stable https://zuul.opendev.org/t/openstack/build/03cf6e83df5848c5b0b4aa4b892806cd : SUCCESS in 1h 20m 36s\n- openstack-ansible-deploy-aio_metal-centos-7 https://zuul.opendev.org/t/openstack/build/92cb70066f4941c3bfd45753f0fd7e18 : SUCCESS in 1h 35m 53s\n- openstack-ansible-deploy-aio_metal-opensuse-15 https://zuul.opendev.org/t/openstack/build/3eff955c2b664137b9214dae3b887676 : SUCCESS in 1h 24m 36s\n- openstack-ansible-deploy-aio_metal-ubuntu-bionic https://zuul.opendev.org/t/openstack/build/f6117ebb23524cd09b440103ddbef0a6 : SUCCESS in 1h 05m 40s\n- openstack-ansible-deploy-aio_distro_metal-centos-7 https://zuul.opendev.org/t/openstack/build/f2e0b42c99c548e49a1f9adaa2e6546c : SUCCESS in 1h 36m 30s\n- openstack-ansible-deploy-aio_distro_metal-opensuse-15 https://zuul.opendev.org/t/openstack/build/a73527c5aa204e6d953252a8f7cf6153 : FAILURE in 46m 09s (non-voting)\n- openstack-ansible-deploy-aio_distro_metal-ubuntu-bionic https://zuul.opendev.org/t/openstack/build/ea03f8b9d4de419a8cdd748856f93acd : SUCCESS in 56m 04s\n- openstack-ansible-deploy-aio_telemetry_metal-centos-7 https://zuul.opendev.org/t/openstack/build/ac7d36e1a82741ccb968781e3a09d8fa : SUCCESS in 1h 37m 09s\n- openstack-ansible-deploy-aio_telemetry_metal-debian-stable https://zuul.opendev.org/t/openstack/build/197e31cd1ba34b02abb1d2eb898b8092 : SUCCESS in 1h 12m 11s\n- openstack-ansible-deploy-aio_telemetry_metal-ubuntu-bionic https://zuul.opendev.org/t/openstack/build/091b335a90774209869bbde68ff4d406 : SUCCESS in 1h 10m 21s\n- openstack-tox-docs https://zuul.opendev.org/t/openstack/build/36a49c5b94ce4973b340ef9dffaf34d2 : SUCCESS in 4m 30s\n- build-openstack-releasenotes https://zuul.opendev.org/t/openstack/build/5fd01eefd3c048efb35fcb378ad70f86 : SUCCESS in 3m 22s","accounts_in_message":[],"_revision_number":5},{"id":"b6d62f3443bf987d49e47e75f3de54180100bfa6","author":{"_account_id":28619,"name":"Dmitriy Rabotyagov","email":"noonedeadpunk@gmail.com","username":"noonedeadpunk"},"date":"2019-08-20 10:54:17.000000000","message":"Patch Set 5: Code-Review+2 Workflow+1","accounts_in_message":[],"_revision_number":5},{"id":"05dec479ba4eb302ec6075aa700729df390e6778","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2019-08-20 10:54:28.000000000","message":"Patch Set 5: -Verified\n\nStarting gate jobs.","accounts_in_message":[],"_revision_number":5},{"id":"bf34c91641a4222d1459c6826cfed7f16943a395","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2019-08-20 12:50:05.000000000","message":"Patch Set 5: Verified-2\n\nBuild failed (gate pipeline).  For information on how to proceed, see\nhttp://docs.openstack.org/infra/manual/developers.html#automated-testing\n\n\n- requirements-check https://zuul.opendev.org/t/openstack/build/bce00b60fae3472ea57ea1b649d7880f : SUCCESS in 4m 06s\n- openstack-ansible-linters https://zuul.opendev.org/t/openstack/build/de80766b90064092ab1b4517c6039eb1 : SUCCESS in 9m 05s\n- openstack-ansible-deploy-aio_metal-debian-stable https://zuul.opendev.org/t/openstack/build/8a46528e585d4208952cd9b128a3dc3e : SUCCESS in 1h 18m 16s\n- openstack-ansible-deploy-aio_metal-centos-7 https://zuul.opendev.org/t/openstack/build/459e6253db8f43eb9fc91c9762f482f4 : SUCCESS in 1h 45m 59s\n- openstack-ansible-deploy-aio_metal-opensuse-15 https://zuul.opendev.org/t/openstack/build/ddb0c8b0bec74cf5ae65c8f21a08688f : FAILURE in 1h 24m 56s\n- openstack-ansible-deploy-aio_metal-ubuntu-bionic https://zuul.opendev.org/t/openstack/build/7c8e785a89cd43b0b4ab53684d7571ed : FAILURE in 1h 26m 17s\n- openstack-ansible-deploy-aio_distro_metal-centos-7 https://zuul.opendev.org/t/openstack/build/8e135ba6d8b34cfa8a55a112de37cb0d : SUCCESS in 1h 14m 04s\n- openstack-ansible-deploy-aio_distro_metal-ubuntu-bionic https://zuul.opendev.org/t/openstack/build/d6cc178148994f278dc27083faf158b9 : SUCCESS in 1h 14m 57s\n- openstack-ansible-deploy-aio_telemetry_metal-centos-7 https://zuul.opendev.org/t/openstack/build/31ad4659c03b40ef97dffe47c3118547 : SUCCESS in 1h 37m 06s\n- openstack-ansible-deploy-aio_telemetry_metal-debian-stable https://zuul.opendev.org/t/openstack/build/f3a4c489a0da4f9080af1532fdd608a0 : SUCCESS in 1h 35m 21s\n- openstack-ansible-deploy-aio_telemetry_metal-ubuntu-bionic https://zuul.opendev.org/t/openstack/build/d1f7458f2626468cb500a4ad4014e70e : SUCCESS in 1h 14m 33s\n- openstack-tox-docs https://zuul.opendev.org/t/openstack/build/42b6cc84eb924f098f1954cceda863fe : SUCCESS in 4m 11s\n- build-openstack-releasenotes https://zuul.opendev.org/t/openstack/build/020438bce3ba49c89103e36ed917506f : SUCCESS in 5m 40s","accounts_in_message":[],"_revision_number":5},{"id":"dd50c55685c56d60fba4654522d1f19e33667cf2","author":{"_account_id":28619,"name":"Dmitriy Rabotyagov","email":"noonedeadpunk@gmail.com","username":"noonedeadpunk"},"date":"2019-08-20 13:20:40.000000000","message":"Patch Set 5:\n\nrecheck","accounts_in_message":[],"_revision_number":5},{"id":"18b9c39eaa38281d88d4e5e961a9c50a9fbc0735","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2019-08-20 15:33:37.000000000","message":"Patch Set 5: Verified+1\n\nBuild succeeded (check pipeline).\n\n- requirements-check https://zuul.opendev.org/t/openstack/build/6eca99d1af0c4ece99ca609a4917d798 : SUCCESS in 4m 23s\n- openstack-ansible-linters https://zuul.opendev.org/t/openstack/build/d8a30e6231a44f8e8c910435e3e75087 : SUCCESS in 4m 55s\n- openstack-ansible-deploy-aio_metal-debian-stable https://zuul.opendev.org/t/openstack/build/17c418e1b7f8482695a66761a2a31a1a : SUCCESS in 1h 22m 07s\n- openstack-ansible-deploy-aio_metal-centos-7 https://zuul.opendev.org/t/openstack/build/fa307a690a1c41ee8f926fa69caf0120 : SUCCESS in 2h 06m 50s\n- openstack-ansible-deploy-aio_metal-opensuse-15 https://zuul.opendev.org/t/openstack/build/c63e36de128d4c1095c9f75fe99ec4fb : SUCCESS in 1h 23m 55s\n- openstack-ansible-deploy-aio_metal-ubuntu-bionic https://zuul.opendev.org/t/openstack/build/33b6e44ea4e84ed99eadb0a9d280f09c : SUCCESS in 1h 15m 23s\n- openstack-ansible-deploy-aio_distro_metal-centos-7 https://zuul.opendev.org/t/openstack/build/5be9ca354a454543a3acef4603811880 : SUCCESS in 1h 16m 35s\n- openstack-ansible-deploy-aio_distro_metal-opensuse-15 https://zuul.opendev.org/t/openstack/build/8e4d7675ab144232beef79e01bb532e8 : FAILURE in 48m 38s (non-voting)\n- openstack-ansible-deploy-aio_distro_metal-ubuntu-bionic https://zuul.opendev.org/t/openstack/build/257c71437e0d44ad87fef364144f57fc : SUCCESS in 1h 06m 51s\n- openstack-ansible-deploy-aio_telemetry_metal-centos-7 https://zuul.opendev.org/t/openstack/build/4aee964b8a6e47309923aa9b0291b036 : SUCCESS in 1h 37m 20s\n- openstack-ansible-deploy-aio_telemetry_metal-debian-stable https://zuul.opendev.org/t/openstack/build/988db900e8174277ad3d9f6ba2bae612 : SUCCESS in 1h 17m 57s\n- openstack-ansible-deploy-aio_telemetry_metal-ubuntu-bionic https://zuul.opendev.org/t/openstack/build/9e0ecbc045704b28910480867bb6524b : SUCCESS in 1h 19m 35s\n- openstack-tox-docs https://zuul.opendev.org/t/openstack/build/51a0495deb7446518d06f395279b1441 : SUCCESS in 4m 16s\n- build-openstack-releasenotes https://zuul.opendev.org/t/openstack/build/a45be760fa084abab4b0bd05d7282454 : SUCCESS in 4m 13s","accounts_in_message":[],"_revision_number":5},{"id":"086b09985e2bc6bb225d25080c2ccc9b9b8a14b4","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2019-08-20 15:33:50.000000000","message":"Patch Set 5: -Verified\n\nStarting gate jobs.","accounts_in_message":[],"_revision_number":5},{"id":"98634ffe688f06c7d1882403e444980b4233dead","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2019-08-20 17:35:18.000000000","message":"Patch Set 5: Verified-2\n\nBuild failed (gate pipeline).  For information on how to proceed, see\nhttp://docs.openstack.org/infra/manual/developers.html#automated-testing\n\n\n- requirements-check https://zuul.opendev.org/t/openstack/build/32c333d3dd424c73858926746070111a : SUCCESS in 3m 20s\n- openstack-ansible-linters https://zuul.opendev.org/t/openstack/build/bdd307809c6d46abbdbc3d13a41ecee6 : SUCCESS in 10m 51s\n- openstack-ansible-deploy-aio_metal-debian-stable https://zuul.opendev.org/t/openstack/build/92b339cf1abc47e6892b31444030d775 : SUCCESS in 1h 17m 23s\n- openstack-ansible-deploy-aio_metal-centos-7 https://zuul.opendev.org/t/openstack/build/1aec5ade553144aaaaec82be7b9814b5 : SUCCESS in 1h 58m 09s\n- openstack-ansible-deploy-aio_metal-opensuse-15 https://zuul.opendev.org/t/openstack/build/316c780879424ef4b1abc751bde28c2c : FAILURE in 1h 23m 38s\n- openstack-ansible-deploy-aio_metal-ubuntu-bionic https://zuul.opendev.org/t/openstack/build/d837e4d1b62e4a64b7f31b90886a716d : FAILURE in 1h 15m 52s\n- openstack-ansible-deploy-aio_distro_metal-centos-7 https://zuul.opendev.org/t/openstack/build/3bba8a0944db4062a4726216d5682037 : SUCCESS in 1h 40m 00s\n- openstack-ansible-deploy-aio_distro_metal-ubuntu-bionic https://zuul.opendev.org/t/openstack/build/ac02b2dc56bb445888b976c71a774b41 : SUCCESS in 1h 00m 39s\n- openstack-ansible-deploy-aio_telemetry_metal-centos-7 https://zuul.opendev.org/t/openstack/build/7d1074cb1c7243ac99064d4ff4c9a9dd : SUCCESS in 1h 58m 01s\n- openstack-ansible-deploy-aio_telemetry_metal-debian-stable https://zuul.opendev.org/t/openstack/build/dbe9190d89a646e7bc4389c707c16855 : SUCCESS in 1h 38m 31s\n- openstack-ansible-deploy-aio_telemetry_metal-ubuntu-bionic https://zuul.opendev.org/t/openstack/build/6bd991297858460a95a2375207b30500 : SUCCESS in 1h 26m 28s\n- openstack-tox-docs https://zuul.opendev.org/t/openstack/build/a72383f2578741a384e3b80c73eb5d39 : SUCCESS in 3m 38s\n- build-openstack-releasenotes https://zuul.opendev.org/t/openstack/build/d18376f6f8024733a1ec82ac56e81407 : SUCCESS in 3m 42s","accounts_in_message":[],"_revision_number":5},{"id":"3b3e4b487b7643aae95f18f4c831c7a16bcd259b","author":{"_account_id":28619,"name":"Dmitriy Rabotyagov","email":"noonedeadpunk@gmail.com","username":"noonedeadpunk"},"date":"2019-08-20 17:43:49.000000000","message":"Patch Set 5:\n\nrecheck","accounts_in_message":[],"_revision_number":5},{"id":"12df2a98d74393cda49a137e37b1f047af4a9d4e","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2019-08-20 19:46:56.000000000","message":"Patch Set 5: Verified+1\n\nBuild succeeded (check pipeline).\n\n- requirements-check https://zuul.opendev.org/t/openstack/build/c2d35931029c4db39e52d31f82aabbb0 : SUCCESS in 3m 53s\n- openstack-ansible-linters https://zuul.opendev.org/t/openstack/build/872560c7a45449dea69ce76d4f6659c3 : SUCCESS in 9m 37s\n- openstack-ansible-deploy-aio_metal-debian-stable https://zuul.opendev.org/t/openstack/build/3cd4d091d60249c7acee517556eb4147 : SUCCESS in 1h 20m 11s\n- openstack-ansible-deploy-aio_metal-centos-7 https://zuul.opendev.org/t/openstack/build/67306e8115104120a1751e6b0eb0ae01 : SUCCESS in 1h 39m 06s\n- openstack-ansible-deploy-aio_metal-opensuse-15 https://zuul.opendev.org/t/openstack/build/6c55260be3a9435d9cc356589d2f2a1d : SUCCESS in 1h 19m 53s\n- openstack-ansible-deploy-aio_metal-ubuntu-bionic https://zuul.opendev.org/t/openstack/build/72fba93dad024b00a35b99c1579d30f9 : SUCCESS in 1h 59m 27s\n- openstack-ansible-deploy-aio_distro_metal-centos-7 https://zuul.opendev.org/t/openstack/build/bb5a0890ac084a7db5a3566f7f4a4c1b : SUCCESS in 1h 22m 21s\n- openstack-ansible-deploy-aio_distro_metal-opensuse-15 https://zuul.opendev.org/t/openstack/build/f0d1e577c4b54435b68d2db9c975f19f : FAILURE in 1h 00m 46s (non-voting)\n- openstack-ansible-deploy-aio_distro_metal-ubuntu-bionic https://zuul.opendev.org/t/openstack/build/4d34dd9880814705b69afe4510af3166 : SUCCESS in 1h 16m 00s\n- openstack-ansible-deploy-aio_telemetry_metal-centos-7 https://zuul.opendev.org/t/openstack/build/e5420ada8e574c3fb8e320beafa7b9d0 : SUCCESS in 1h 51m 50s\n- openstack-ansible-deploy-aio_telemetry_metal-debian-stable https://zuul.opendev.org/t/openstack/build/ba44f0ebe83f423ea05596fb5ba0e5c3 : SUCCESS in 1h 18m 04s\n- openstack-ansible-deploy-aio_telemetry_metal-ubuntu-bionic https://zuul.opendev.org/t/openstack/build/56b2bdbecd3840759412d63c9f3da4dc : SUCCESS in 1h 26m 05s\n- openstack-tox-docs https://zuul.opendev.org/t/openstack/build/95f88ed7b2694909ac95c3e6e2054d26 : SUCCESS in 3m 40s\n- build-openstack-releasenotes https://zuul.opendev.org/t/openstack/build/598afd7702cf4cfe9eb083eab1898467 : SUCCESS in 3m 45s","accounts_in_message":[],"_revision_number":5},{"id":"5e5ba2543e94f61d99184b21a7778e1cd973c01d","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2019-08-20 19:47:12.000000000","message":"Patch Set 5: -Verified\n\nStarting gate jobs.","accounts_in_message":[],"_revision_number":5},{"id":"50e8681b0eb1f4007087d45a7f696211b44833b6","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2019-08-20 22:03:24.000000000","message":"Patch Set 5: Verified-2\n\nBuild failed (gate pipeline).  For information on how to proceed, see\nhttp://docs.openstack.org/infra/manual/developers.html#automated-testing\n\n\n- requirements-check https://zuul.opendev.org/t/openstack/build/784e074a7cb148568cad6d2aa2698fad : SUCCESS in 3m 23s\n- openstack-ansible-linters https://zuul.opendev.org/t/openstack/build/720e9568fde14e5ebe3aeea62f2f9358 : SUCCESS in 6m 19s\n- openstack-ansible-deploy-aio_metal-debian-stable https://zuul.opendev.org/t/openstack/build/33fa00b97a2e4cab9e71a93f247ddc3f : SUCCESS in 1h 16m 12s\n- openstack-ansible-deploy-aio_metal-centos-7 https://zuul.opendev.org/t/openstack/build/ae439b93b4404705901be7f4a312d9eb : SUCCESS in 1h 55m 51s\n- openstack-ansible-deploy-aio_metal-opensuse-15 https://zuul.opendev.org/t/openstack/build/617aec1bf37b4745b4af25f2c93ee92d : SUCCESS in 1h 21m 01s\n- openstack-ansible-deploy-aio_metal-ubuntu-bionic https://zuul.opendev.org/t/openstack/build/b68105a687284b9a8e1c9af91396e45c : FAILURE in 1h 30m 30s\n- openstack-ansible-deploy-aio_distro_metal-centos-7 https://zuul.opendev.org/t/openstack/build/310cd559c2504917820d73e680c42e89 : SUCCESS in 1h 09m 55s\n- openstack-ansible-deploy-aio_distro_metal-ubuntu-bionic https://zuul.opendev.org/t/openstack/build/3f1f9bee6cd749e3b91e08a4bb8d567c : SUCCESS in 1h 12m 45s\n- openstack-ansible-deploy-aio_telemetry_metal-centos-7 https://zuul.opendev.org/t/openstack/build/5f2aff00798144fb9d93d2cd1b7b7e5d : SUCCESS in 1h 44m 31s\n- openstack-ansible-deploy-aio_telemetry_metal-debian-stable https://zuul.opendev.org/t/openstack/build/dc3947d957cb4231b2acd7216eef6b7a : SUCCESS in 1h 27m 23s\n- openstack-ansible-deploy-aio_telemetry_metal-ubuntu-bionic https://zuul.opendev.org/t/openstack/build/99792b6d665c4488b7a0d30c8b6e2e5b : SUCCESS in 1h 12m 29s\n- openstack-tox-docs https://zuul.opendev.org/t/openstack/build/9ccbceca0991449a9c20b5c42750e21b : SUCCESS in 3m 15s\n- build-openstack-releasenotes https://zuul.opendev.org/t/openstack/build/1a5fa323110d4330aee454b724638fa8 : SUCCESS in 4m 38s","accounts_in_message":[],"_revision_number":5},{"id":"79836912953bf5ae0be5f863b74bee8dc80bb0fd","author":{"_account_id":28619,"name":"Dmitriy Rabotyagov","email":"noonedeadpunk@gmail.com","username":"noonedeadpunk"},"date":"2019-08-21 10:57:54.000000000","message":"Patch Set 5:\n\nrecheck","accounts_in_message":[],"_revision_number":5},{"id":"99d6b7c124a338bba466952c8c50f493488d7089","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2019-08-21 13:01:31.000000000","message":"Patch Set 5: Verified+1\n\nBuild succeeded (check pipeline).\n\n- requirements-check https://zuul.opendev.org/t/openstack/build/9047b8047a964e6bb385349d5ac25ee9 : SUCCESS in 2m 57s\n- openstack-ansible-linters https://zuul.opendev.org/t/openstack/build/ac8aada8e6c942da82602157b700c406 : SUCCESS in 7m 58s\n- openstack-ansible-deploy-aio_metal-debian-stable https://zuul.opendev.org/t/openstack/build/78c1c602aaf14bc8a7785dab1ed71e14 : SUCCESS in 1h 16m 49s\n- openstack-ansible-deploy-aio_metal-centos-7 https://zuul.opendev.org/t/openstack/build/caf3ce279cd748789a3caf116826900b : SUCCESS in 1h 56m 47s\n- openstack-ansible-deploy-aio_metal-opensuse-15 https://zuul.opendev.org/t/openstack/build/db1d5c21ad1945b89a5961dba432828c : SUCCESS in 1h 23m 22s\n- openstack-ansible-deploy-aio_metal-ubuntu-bionic https://zuul.opendev.org/t/openstack/build/a617f68939e74fb4b106cda0cf40500d : SUCCESS in 1h 13m 17s\n- openstack-ansible-deploy-aio_distro_metal-centos-7 https://zuul.opendev.org/t/openstack/build/5fd71b79e9554f00adf52df42e3bb1af : SUCCESS in 1h 32m 05s\n- openstack-ansible-deploy-aio_distro_metal-opensuse-15 https://zuul.opendev.org/t/openstack/build/03034d7968e54cd0a175c2d0c16fdd8b : FAILURE in 46m 26s (non-voting)\n- openstack-ansible-deploy-aio_distro_metal-ubuntu-bionic https://zuul.opendev.org/t/openstack/build/35412c7da48045b7a1a484740ed64cab : SUCCESS in 54m 40s\n- openstack-ansible-deploy-aio_telemetry_metal-centos-7 https://zuul.opendev.org/t/openstack/build/23b4d08acee142eab514675c9802ac93 : SUCCESS in 1h 54m 37s\n- openstack-ansible-deploy-aio_telemetry_metal-debian-stable https://zuul.opendev.org/t/openstack/build/4e670b61cb204cd3af760188d4cea13a : SUCCESS in 1h 52m 06s\n- openstack-ansible-deploy-aio_telemetry_metal-ubuntu-bionic https://zuul.opendev.org/t/openstack/build/d91174708fb041bd8d325373c45b694f : SUCCESS in 1h 14m 11s\n- openstack-tox-docs https://zuul.opendev.org/t/openstack/build/66aec36ed0e94e8ea995fd2712d3ca8f : SUCCESS in 3m 45s\n- build-openstack-releasenotes https://zuul.opendev.org/t/openstack/build/cfc0b6571f6c42c9aecdc738c802dd79 : SUCCESS in 4m 39s","accounts_in_message":[],"_revision_number":5},{"id":"c5f8d13a0a5a985172dad69052fe6bfcf297adcc","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2019-08-21 13:01:51.000000000","message":"Patch Set 5: -Verified\n\nStarting gate jobs.","accounts_in_message":[],"_revision_number":5},{"id":"518fc2c6b12f9d589e32507a114de17a8e6b184f","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2019-08-21 15:00:11.000000000","message":"Patch Set 5: Verified-2\n\nBuild failed (gate pipeline).  For information on how to proceed, see\nhttp://docs.openstack.org/infra/manual/developers.html#automated-testing\n\n\n- requirements-check https://zuul.opendev.org/t/openstack/build/eba23b257df7491fa94c9fd91753d4ca : SUCCESS in 2m 27s\n- openstack-ansible-linters https://zuul.opendev.org/t/openstack/build/8c0bf943ee144baabd9bdbadd0326798 : SUCCESS in 6m 49s\n- openstack-ansible-deploy-aio_metal-debian-stable https://zuul.opendev.org/t/openstack/build/c75d2782426f4749beba132262916e40 : FAILURE in 46m 22s\n- openstack-ansible-deploy-aio_metal-centos-7 https://zuul.opendev.org/t/openstack/build/c8132b69392f4e88879024a1c564c8f4 : SUCCESS in 1h 44m 25s\n- openstack-ansible-deploy-aio_metal-opensuse-15 https://zuul.opendev.org/t/openstack/build/974827b424d243b996f73fae20383d90 : FAILURE in 1h 34m 29s\n- openstack-ansible-deploy-aio_metal-ubuntu-bionic https://zuul.opendev.org/t/openstack/build/56e7667b36294dcfaf0c987c8dcc3bb5 : FAILURE in 46m 27s\n- openstack-ansible-deploy-aio_distro_metal-centos-7 https://zuul.opendev.org/t/openstack/build/3241f4554a8c46d6a6ffa0966387f155 : SUCCESS in 1h 20m 06s\n- openstack-ansible-deploy-aio_distro_metal-ubuntu-bionic https://zuul.opendev.org/t/openstack/build/989c568eea784ee1915831603e2cf0d2 : FAILURE in 52m 21s\n- openstack-ansible-deploy-aio_telemetry_metal-centos-7 https://zuul.opendev.org/t/openstack/build/d96aa27eeb064759936577f96e01cf68 : SUCCESS in 1h 46m 37s\n- openstack-ansible-deploy-aio_telemetry_metal-debian-stable https://zuul.opendev.org/t/openstack/build/d35cef26a57f45f19592c95f25f74d63 : FAILURE in 45m 02s\n- openstack-ansible-deploy-aio_telemetry_metal-ubuntu-bionic https://zuul.opendev.org/t/openstack/build/dcbb9325bb844099a9b2fbbe823342d2 : FAILURE in 50m 28s\n- openstack-tox-docs https://zuul.opendev.org/t/openstack/build/c200f8b8790645288dbe53dc0a4bd1da : SUCCESS in 3m 43s\n- build-openstack-releasenotes https://zuul.opendev.org/t/openstack/build/04c7ff476f3d4303aeaa8fbc9bb132b4 : SUCCESS in 3m 49s","accounts_in_message":[],"_revision_number":5},{"id":"a2cd6eb23ccfeb689fab50269f5581760d60967a","author":{"_account_id":28619,"name":"Dmitriy Rabotyagov","email":"noonedeadpunk@gmail.com","username":"noonedeadpunk"},"date":"2019-08-23 02:01:09.000000000","message":"Patch Set 5:\n\nrecheck","accounts_in_message":[],"_revision_number":5},{"id":"3ceee43f3a628b7ace86fc2a9f0247d0fd98a244","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2019-08-23 04:23:09.000000000","message":"Patch Set 5: Verified-1\n\nBuild failed (check pipeline).  For information on how to proceed, see\nhttp://docs.openstack.org/infra/manual/developers.html#automated-testing\n\n\n- requirements-check https://zuul.opendev.org/t/openstack/build/4f85f3bc17134314afc63792454d94b5 : SUCCESS in 3m 10s\n- openstack-ansible-linters https://zuul.opendev.org/t/openstack/build/c43e7023365842a6bdc643df39bb51d6 : SUCCESS in 4m 44s\n- openstack-ansible-deploy-aio_metal-debian-stable https://zuul.opendev.org/t/openstack/build/06b74a3813864d6b939fc4638a58a8aa : SUCCESS in 1h 34m 02s\n- openstack-ansible-deploy-aio_metal-centos-7 https://zuul.opendev.org/t/openstack/build/dd2eb0bc2a624353b35dd23a20da27ba : SUCCESS in 2h 20m 21s\n- openstack-ansible-deploy-aio_metal-opensuse-15 https://zuul.opendev.org/t/openstack/build/4d83ad6168eb420dbb7e9908a97ffade : SUCCESS in 1h 49m 23s\n- openstack-ansible-deploy-aio_metal-ubuntu-bionic https://zuul.opendev.org/t/openstack/build/12e563250986437a944b84bd856f4d57 : SUCCESS in 1h 22m 43s\n- openstack-ansible-deploy-aio_distro_metal-centos-7 https://zuul.opendev.org/t/openstack/build/3c708478b4dd4b7eaa493e81728ad29f : SUCCESS in 1h 24m 07s\n- openstack-ansible-deploy-aio_distro_metal-opensuse-15 https://zuul.opendev.org/t/openstack/build/597b4d374c8344b0a93074f1c616e126 : FAILURE in 59m 44s (non-voting)\n- openstack-ansible-deploy-aio_distro_metal-ubuntu-bionic https://zuul.opendev.org/t/openstack/build/f04afdb831584605a5f716940748d38a : SUCCESS in 58m 39s\n- openstack-ansible-deploy-aio_telemetry_metal-centos-7 https://zuul.opendev.org/t/openstack/build/31a6fd2e8e9d43bab9efba8d21a6b7c6 : SUCCESS in 1h 48m 36s\n- openstack-ansible-deploy-aio_telemetry_metal-debian-stable https://zuul.opendev.org/t/openstack/build/3fa68eabcb9b4732b768f9b69fb2e204 : SUCCESS in 1h 29m 58s\n- openstack-ansible-deploy-aio_telemetry_metal-ubuntu-bionic https://zuul.opendev.org/t/openstack/build/6babc458ba2542b3b255aff30982119e : FAILURE in 1h 26m 14s\n- openstack-tox-docs https://zuul.opendev.org/t/openstack/build/8f315237a092476484d762b2becbeb11 : SUCCESS in 5m 46s\n- build-openstack-releasenotes https://zuul.opendev.org/t/openstack/build/f328189b608a47658e33da169921d221 : SUCCESS in 3m 52s","accounts_in_message":[],"_revision_number":5},{"id":"efad8d307ebb182ce06a51e60b10353399acbc51","author":{"_account_id":28619,"name":"Dmitriy Rabotyagov","email":"noonedeadpunk@gmail.com","username":"noonedeadpunk"},"date":"2019-08-23 06:26:14.000000000","message":"Patch Set 5:\n\nrecheck","accounts_in_message":[],"_revision_number":5},{"id":"eef3560ab77d472eca33de5704c0cee7eaeba5de","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2019-08-23 08:52:11.000000000","message":"Patch Set 5: Verified+1\n\nBuild succeeded (check pipeline).\n\n- requirements-check https://zuul.opendev.org/t/openstack/build/0c4c84f923464ee8b476b2278cf04c2f : SUCCESS in 3m 34s\n- openstack-ansible-linters https://zuul.opendev.org/t/openstack/build/812aa1d1d3d74a6ca3714e41cf4f404a : SUCCESS in 6m 58s\n- openstack-ansible-deploy-aio_metal-debian-stable https://zuul.opendev.org/t/openstack/build/1557770d77c4483897cc74cef27a158b : SUCCESS in 1h 35m 24s\n- openstack-ansible-deploy-aio_metal-centos-7 https://zuul.opendev.org/t/openstack/build/e2b94dc7b9e04deab35490f601438b60 : SUCCESS in 2h 16m 49s\n- openstack-ansible-deploy-aio_metal-opensuse-15 https://zuul.opendev.org/t/openstack/build/88a5fd8d18bf4c5eb6bddf32bdf4e95b : SUCCESS in 1h 42m 04s\n- openstack-ansible-deploy-aio_metal-ubuntu-bionic https://zuul.opendev.org/t/openstack/build/2b5099deb9ba42a7ab37a658a096006f : SUCCESS in 1h 25m 03s\n- openstack-ansible-deploy-aio_distro_metal-centos-7 https://zuul.opendev.org/t/openstack/build/55c7a22f21b548a6a4fd5316d8623022 : SUCCESS in 1h 24m 42s\n- openstack-ansible-deploy-aio_distro_metal-opensuse-15 https://zuul.opendev.org/t/openstack/build/d551eb33a33841529b099d5bf09449ba : FAILURE in 58m 02s (non-voting)\n- openstack-ansible-deploy-aio_distro_metal-ubuntu-bionic https://zuul.opendev.org/t/openstack/build/e9138e2ae8514c5f82736a2c58275b20 : SUCCESS in 57m 31s\n- openstack-ansible-deploy-aio_telemetry_metal-centos-7 https://zuul.opendev.org/t/openstack/build/d9861a22aeb64e419d92d6c7f040feab : SUCCESS in 1h 42m 40s\n- openstack-ansible-deploy-aio_telemetry_metal-debian-stable https://zuul.opendev.org/t/openstack/build/20d69f38d987438abfcab5efe1bafcd2 : SUCCESS in 1h 27m 10s\n- openstack-ansible-deploy-aio_telemetry_metal-ubuntu-bionic https://zuul.opendev.org/t/openstack/build/c37d319efc7d4cf2beecab7c46565de5 : SUCCESS in 1h 23m 31s\n- openstack-tox-docs https://zuul.opendev.org/t/openstack/build/8a2d884b44b84c2f88127b3ed4e03755 : SUCCESS in 3m 43s\n- build-openstack-releasenotes https://zuul.opendev.org/t/openstack/build/2b7b9c8a74a446db9c8777d618254fea : SUCCESS in 3m 48s","accounts_in_message":[],"_revision_number":5},{"id":"915f6f4e514f6a4aeab65de7a00286a14803a0d4","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2019-08-23 08:52:22.000000000","message":"Patch Set 5: -Verified\n\nStarting gate jobs.","accounts_in_message":[],"_revision_number":5},{"id":"f1ca919fb02536798e570a956b611e1b75d89a69","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2019-08-23 10:53:30.000000000","message":"Change has been successfully merged by Zuul","accounts_in_message":[],"_revision_number":5},{"id":"103fab6847b4124d1882237677bfefcba548ded2","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2019-08-23 10:53:30.000000000","message":"Patch Set 5: Verified+2\n\nBuild succeeded (gate pipeline).\n\n- requirements-check https://zuul.opendev.org/t/openstack/build/1c9fd7843582434491dce97414037c70 : SUCCESS in 2m 50s\n- openstack-ansible-linters https://zuul.opendev.org/t/openstack/build/ea26e13ab3b546658a9a2f5ed076efbc : SUCCESS in 7m 25s\n- openstack-ansible-deploy-aio_metal-debian-stable https://zuul.opendev.org/t/openstack/build/7b7e85493e5641c191b66645ce9a668d : SUCCESS in 1h 15m 01s\n- openstack-ansible-deploy-aio_metal-centos-7 https://zuul.opendev.org/t/openstack/build/f873166efe6f4c2f903194b1b2d93759 : SUCCESS in 1h 45m 03s\n- openstack-ansible-deploy-aio_metal-opensuse-15 https://zuul.opendev.org/t/openstack/build/b49fbe0fbfb948a285091adf2e6724e6 : SUCCESS in 1h 28m 24s\n- openstack-ansible-deploy-aio_metal-ubuntu-bionic https://zuul.opendev.org/t/openstack/build/fec8d5af5970445c93598780351ae6ce : SUCCESS in 1h 22m 36s\n- openstack-ansible-deploy-aio_distro_metal-centos-7 https://zuul.opendev.org/t/openstack/build/7ba1a3c971e645b189a5ab0dc4ed9474 : SUCCESS in 1h 31m 54s\n- openstack-ansible-deploy-aio_distro_metal-ubuntu-bionic https://zuul.opendev.org/t/openstack/build/00949e095349475a88afc931fa36d2be : SUCCESS in 1h 02m 37s\n- openstack-ansible-deploy-aio_telemetry_metal-centos-7 https://zuul.opendev.org/t/openstack/build/faf863e94cb149a2b82971450f13a335 : SUCCESS in 1h 55m 29s\n- openstack-ansible-deploy-aio_telemetry_metal-debian-stable https://zuul.opendev.org/t/openstack/build/a61ee7125ad84edf86b8d4ddc76ca803 : SUCCESS in 1h 16m 51s\n- openstack-ansible-deploy-aio_telemetry_metal-ubuntu-bionic https://zuul.opendev.org/t/openstack/build/05c86d391b204476aa330a1eb8047b4f : SUCCESS in 1h 16m 58s\n- openstack-tox-docs https://zuul.opendev.org/t/openstack/build/c2e76a414d7e452fa5d03ed438eb3f14 : SUCCESS in 4m 17s\n- build-openstack-releasenotes https://zuul.opendev.org/t/openstack/build/03ec252afdc84653ba7f2d90b45bd7ff : SUCCESS in 5m 12s","accounts_in_message":[],"_revision_number":5},{"id":"e65c83261154677b720747d9f1440b79f5c1d4c3","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2019-08-23 10:55:10.000000000","message":"Patch Set 5:\n\nBuild succeeded (promote pipeline).\n\n- promote-openstack-tox-docs https://zuul.opendev.org/t/openstack/build/1c9f94232cea4b37a6cdf9214a5f296e : SUCCESS in 1m 22s","accounts_in_message":[],"_revision_number":5}],"current_revision_number":5,"current_revision":"d4422a137d2323eeddb4e278e288be3428f91011","revisions":{"7fcea0a96e1db62fba62b1541045acbdc73897b5":{"kind":"REWORK","_number":1,"created":"2019-07-18 22:17:32.000000000","uploader":{"_account_id":11131,"name":"OpenStack Proposal Bot","username":"proposal-bot","tags":["SERVICE_USER"]},"ref":"refs/changes/90/671590/1","fetch":{"anonymous http":{"url":"https://review.opendev.org/openstack/openstack-ansible-os_aodh","ref":"refs/changes/90/671590/1","commands":{"Checkout":"git fetch https://review.opendev.org/openstack/openstack-ansible-os_aodh refs/changes/90/671590/1 \u0026\u0026 git checkout FETCH_HEAD","Cherry Pick":"git fetch https://review.opendev.org/openstack/openstack-ansible-os_aodh refs/changes/90/671590/1 \u0026\u0026 git cherry-pick FETCH_HEAD","Format Patch":"git fetch https://review.opendev.org/openstack/openstack-ansible-os_aodh refs/changes/90/671590/1 \u0026\u0026 git format-patch -1 --stdout FETCH_HEAD","Pull":"git pull https://review.opendev.org/openstack/openstack-ansible-os_aodh refs/changes/90/671590/1"}}},"commit":{"parents":[{"commit":"cead1225150b333d65862e322633ced095a45b04","subject":"Merge \"Replace git.openstack.org URLs with opendev.org URLs\"","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/openstack-ansible-os_aodh/commit/cead1225150b333d65862e322633ced095a45b04"}]}],"author":{"name":"OpenStack Proposal Bot","email":"openstack-infra@lists.openstack.org","date":"2019-07-18 22:17:30.000000000","tz":0},"committer":{"name":"OpenStack Proposal Bot","email":"openstack-infra@lists.openstack.org","date":"2019-07-18 22:17:30.000000000","tz":0},"subject":"Updated from OpenStack Ansible Tests","message":"Updated from OpenStack Ansible Tests\n\nChange-Id: I407267fc32a4a755ab2ee91095ddc401f68a1b19\n","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/openstack-ansible-os_aodh/commit/7fcea0a96e1db62fba62b1541045acbdc73897b5"}],"resolve_conflicts_web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/openstack-ansible-os_aodh/commit/7fcea0a96e1db62fba62b1541045acbdc73897b5"}]},"branch":"refs/heads/master"},"e4033cfa11f4c39ff1a2e58fb46e1ee6aa9fe93b":{"kind":"REWORK","_number":2,"created":"2019-07-19 18:19:14.000000000","uploader":{"_account_id":11131,"name":"OpenStack Proposal Bot","username":"proposal-bot","tags":["SERVICE_USER"]},"ref":"refs/changes/90/671590/2","fetch":{"anonymous http":{"url":"https://review.opendev.org/openstack/openstack-ansible-os_aodh","ref":"refs/changes/90/671590/2","commands":{"Checkout":"git fetch https://review.opendev.org/openstack/openstack-ansible-os_aodh refs/changes/90/671590/2 \u0026\u0026 git checkout FETCH_HEAD","Cherry Pick":"git fetch https://review.opendev.org/openstack/openstack-ansible-os_aodh refs/changes/90/671590/2 \u0026\u0026 git cherry-pick FETCH_HEAD","Format Patch":"git fetch https://review.opendev.org/openstack/openstack-ansible-os_aodh refs/changes/90/671590/2 \u0026\u0026 git format-patch -1 --stdout FETCH_HEAD","Pull":"git pull https://review.opendev.org/openstack/openstack-ansible-os_aodh refs/changes/90/671590/2"}}},"commit":{"parents":[{"commit":"cead1225150b333d65862e322633ced095a45b04","subject":"Merge \"Replace git.openstack.org URLs with opendev.org URLs\"","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/openstack-ansible-os_aodh/commit/cead1225150b333d65862e322633ced095a45b04"}]}],"author":{"name":"OpenStack Proposal Bot","email":"openstack-infra@lists.openstack.org","date":"2019-07-19 18:19:10.000000000","tz":0},"committer":{"name":"OpenStack Proposal Bot","email":"openstack-infra@lists.openstack.org","date":"2019-07-19 18:19:10.000000000","tz":0},"subject":"Updated from OpenStack Ansible Tests","message":"Updated from OpenStack Ansible Tests\n\nChange-Id: I407267fc32a4a755ab2ee91095ddc401f68a1b19\n","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/openstack-ansible-os_aodh/commit/e4033cfa11f4c39ff1a2e58fb46e1ee6aa9fe93b"}],"resolve_conflicts_web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/openstack-ansible-os_aodh/commit/e4033cfa11f4c39ff1a2e58fb46e1ee6aa9fe93b"}]},"branch":"refs/heads/master"},"24a88129dc01861cfb0f4c9534bbd5e39f5e6989":{"kind":"REWORK","_number":3,"created":"2019-07-20 09:29:46.000000000","uploader":{"_account_id":11131,"name":"OpenStack Proposal Bot","username":"proposal-bot","tags":["SERVICE_USER"]},"ref":"refs/changes/90/671590/3","fetch":{"anonymous http":{"url":"https://review.opendev.org/openstack/openstack-ansible-os_aodh","ref":"refs/changes/90/671590/3","commands":{"Checkout":"git fetch https://review.opendev.org/openstack/openstack-ansible-os_aodh refs/changes/90/671590/3 \u0026\u0026 git checkout FETCH_HEAD","Cherry Pick":"git fetch https://review.opendev.org/openstack/openstack-ansible-os_aodh refs/changes/90/671590/3 \u0026\u0026 git cherry-pick FETCH_HEAD","Format Patch":"git fetch https://review.opendev.org/openstack/openstack-ansible-os_aodh refs/changes/90/671590/3 \u0026\u0026 git format-patch -1 --stdout FETCH_HEAD","Pull":"git pull https://review.opendev.org/openstack/openstack-ansible-os_aodh refs/changes/90/671590/3"}}},"commit":{"parents":[{"commit":"cead1225150b333d65862e322633ced095a45b04","subject":"Merge \"Replace git.openstack.org URLs with opendev.org URLs\"","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/openstack-ansible-os_aodh/commit/cead1225150b333d65862e322633ced095a45b04"}]}],"author":{"name":"OpenStack Proposal Bot","email":"openstack-infra@lists.openstack.org","date":"2019-07-20 09:29:41.000000000","tz":0},"committer":{"name":"OpenStack Proposal Bot","email":"openstack-infra@lists.openstack.org","date":"2019-07-20 09:29:41.000000000","tz":0},"subject":"Updated from OpenStack Ansible Tests","message":"Updated from OpenStack Ansible Tests\n\nChange-Id: I407267fc32a4a755ab2ee91095ddc401f68a1b19\n","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/openstack-ansible-os_aodh/commit/24a88129dc01861cfb0f4c9534bbd5e39f5e6989"}],"resolve_conflicts_web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/openstack-ansible-os_aodh/commit/24a88129dc01861cfb0f4c9534bbd5e39f5e6989"}]},"branch":"refs/heads/master"},"3a987d98ec5dc5cda9fbea804284ed5dadb96a8a":{"kind":"REWORK","_number":4,"created":"2019-07-30 19:37:12.000000000","uploader":{"_account_id":11131,"name":"OpenStack Proposal Bot","username":"proposal-bot","tags":["SERVICE_USER"]},"ref":"refs/changes/90/671590/4","fetch":{"anonymous http":{"url":"https://review.opendev.org/openstack/openstack-ansible-os_aodh","ref":"refs/changes/90/671590/4","commands":{"Checkout":"git fetch https://review.opendev.org/openstack/openstack-ansible-os_aodh refs/changes/90/671590/4 \u0026\u0026 git checkout FETCH_HEAD","Cherry Pick":"git fetch https://review.opendev.org/openstack/openstack-ansible-os_aodh refs/changes/90/671590/4 \u0026\u0026 git cherry-pick FETCH_HEAD","Format Patch":"git fetch https://review.opendev.org/openstack/openstack-ansible-os_aodh refs/changes/90/671590/4 \u0026\u0026 git format-patch -1 --stdout FETCH_HEAD","Pull":"git pull https://review.opendev.org/openstack/openstack-ansible-os_aodh refs/changes/90/671590/4"}}},"commit":{"parents":[{"commit":"cead1225150b333d65862e322633ced095a45b04","subject":"Merge \"Replace git.openstack.org URLs with opendev.org URLs\"","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/openstack-ansible-os_aodh/commit/cead1225150b333d65862e322633ced095a45b04"}]}],"author":{"name":"OpenStack Proposal Bot","email":"openstack-infra@lists.openstack.org","date":"2019-07-30 19:37:08.000000000","tz":0},"committer":{"name":"OpenStack Proposal Bot","email":"openstack-infra@lists.openstack.org","date":"2019-07-30 19:37:08.000000000","tz":0},"subject":"Updated from OpenStack Ansible Tests","message":"Updated from OpenStack Ansible Tests\n\nChange-Id: I407267fc32a4a755ab2ee91095ddc401f68a1b19\n","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/openstack-ansible-os_aodh/commit/3a987d98ec5dc5cda9fbea804284ed5dadb96a8a"}],"resolve_conflicts_web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/openstack-ansible-os_aodh/commit/3a987d98ec5dc5cda9fbea804284ed5dadb96a8a"}]},"branch":"refs/heads/master"},"d4422a137d2323eeddb4e278e288be3428f91011":{"kind":"REWORK","_number":5,"created":"2019-08-20 03:05:33.000000000","uploader":{"_account_id":11131,"name":"OpenStack Proposal Bot","username":"proposal-bot","tags":["SERVICE_USER"]},"ref":"refs/changes/90/671590/5","fetch":{"anonymous http":{"url":"https://review.opendev.org/openstack/openstack-ansible-os_aodh","ref":"refs/changes/90/671590/5","commands":{"Checkout":"git fetch https://review.opendev.org/openstack/openstack-ansible-os_aodh refs/changes/90/671590/5 \u0026\u0026 git checkout FETCH_HEAD","Cherry Pick":"git fetch https://review.opendev.org/openstack/openstack-ansible-os_aodh refs/changes/90/671590/5 \u0026\u0026 git cherry-pick FETCH_HEAD","Format Patch":"git fetch https://review.opendev.org/openstack/openstack-ansible-os_aodh refs/changes/90/671590/5 \u0026\u0026 git format-patch -1 --stdout FETCH_HEAD","Pull":"git pull https://review.opendev.org/openstack/openstack-ansible-os_aodh refs/changes/90/671590/5"}}},"commit":{"parents":[{"commit":"97c03b8decb175176299e6fceb3e275323c0f08a","subject":"service_setup: refactor service setup to a single file","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/openstack-ansible-os_aodh/commit/97c03b8decb175176299e6fceb3e275323c0f08a"}]}],"author":{"name":"OpenStack Proposal Bot","email":"openstack-infra@lists.openstack.org","date":"2019-08-20 03:05:32.000000000","tz":0},"committer":{"name":"OpenStack Proposal Bot","email":"openstack-infra@lists.openstack.org","date":"2019-08-20 03:05:32.000000000","tz":0},"subject":"Updated from OpenStack Ansible Tests","message":"Updated from OpenStack Ansible Tests\n\nChange-Id: I407267fc32a4a755ab2ee91095ddc401f68a1b19\n","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/openstack-ansible-os_aodh/commit/d4422a137d2323eeddb4e278e288be3428f91011"}],"resolve_conflicts_web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/openstack-ansible-os_aodh/commit/d4422a137d2323eeddb4e278e288be3428f91011"}]},"branch":"refs/heads/master"}},"requirements":[],"submit_records":[],"submit_requirements":[]}
