)]}'
{"id":"openstack%2Fopenstack-ansible-os_aodh~677130","triplet_id":"openstack%2Fopenstack-ansible-os_aodh~stable%2Fpike~I9d28aecb323ffac57bb062dea46f27fc545a2b02","project":"openstack/openstack-ansible-os_aodh","branch":"stable/pike","topic":"goodbye_42.3","hashtags":[],"change_id":"I9d28aecb323ffac57bb062dea46f27fc545a2b02","subject":"Remove openSUSE 42.3 job","status":"MERGED","created":"2019-08-19 07:42:19.000000000","updated":"2019-08-27 13:56:59.000000000","submitted":"2019-08-27 13:56:58.000000000","submitter":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"total_comment_count":0,"unresolved_comment_count":0,"has_review_started":true,"submission_id":"677130-1566914219322-1d226e13","meta_rev_id":"dafa5cb7465db2750a2b07c65daf68bbd15af6bf","_number":677130,"virtual_id_number":677130,"owner":{"_account_id":6547,"name":"Andreas Jaeger","email":"jaegerandi@gmail.com","username":"jaegerandi"},"actions":{},"labels":{"Verified":{"approved":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"all":[{"value":0,"date":"2019-08-26 20:01:26.000000000","_account_id":6547,"name":"Andreas Jaeger","email":"jaegerandi@gmail.com","username":"jaegerandi"},{"value":0,"_account_id":28619,"name":"Dmitriy Rabotyagov","email":"noonedeadpunk@gmail.com","username":"noonedeadpunk"},{"value":0,"_account_id":1004,"name":"Mohammed Naser","email":"mnaser@vexxhost.com","username":"mnaser"},{"value":2,"date":"2019-08-27 13:56:58.000000000","permitted_voting_range":{"min":2,"max":2},"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]}],"values":{"-2":"Fails","-1":"Doesn\u0027t seem to work"," 0":"No score","+1":"Works for me","+2":"Verified"},"description":"","default_value":0,"optional":true},"Code-Review":{"approved":{"_account_id":28619,"name":"Dmitriy Rabotyagov","email":"noonedeadpunk@gmail.com","username":"noonedeadpunk"},"all":[{"value":0,"_account_id":6547,"name":"Andreas Jaeger","email":"jaegerandi@gmail.com","username":"jaegerandi"},{"value":2,"date":"2019-08-27 12:35:39.000000000","permitted_voting_range":{"min":2,"max":2},"_account_id":28619,"name":"Dmitriy Rabotyagov","email":"noonedeadpunk@gmail.com","username":"noonedeadpunk"},{"value":2,"date":"2019-08-27 13:01:39.000000000","permitted_voting_range":{"min":2,"max":2},"_account_id":1004,"name":"Mohammed Naser","email":"mnaser@vexxhost.com","username":"mnaser"},{"value":0,"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]}],"values":{"-2":"Do not merge","-1":"This patch needs further work before it can be merged"," 0":"No score","+1":"Looks good to me, but someone else must approve","+2":"Looks good to me (core reviewer)"},"description":"","default_value":0,"optional":true},"Workflow":{"approved":{"_account_id":1004,"name":"Mohammed Naser","email":"mnaser@vexxhost.com","username":"mnaser"},"all":[{"value":0,"_account_id":6547,"name":"Andreas Jaeger","email":"jaegerandi@gmail.com","username":"jaegerandi"},{"value":0,"_account_id":28619,"name":"Dmitriy Rabotyagov","email":"noonedeadpunk@gmail.com","username":"noonedeadpunk"},{"value":1,"date":"2019-08-27 13:01:39.000000000","permitted_voting_range":{"min":1,"max":1},"_account_id":1004,"name":"Mohammed Naser","email":"mnaser@vexxhost.com","username":"mnaser"},{"value":0,"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]}],"values":{"-1":"Work in progress"," 0":"Ready for reviews","+1":"Approved"},"description":"","default_value":0,"optional":true},"Backport-Candidate":{"all":[{"value":0,"_account_id":6547,"name":"Andreas Jaeger","email":"jaegerandi@gmail.com","username":"jaegerandi"},{"value":0,"_account_id":28619,"name":"Dmitriy Rabotyagov","email":"noonedeadpunk@gmail.com","username":"noonedeadpunk"},{"value":0,"_account_id":1004,"name":"Mohammed Naser","email":"mnaser@vexxhost.com","username":"mnaser"},{"value":0,"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]}],"values":{"-1":"Not A Backport Candidate"," 0":"Backport Review Needed","+1":"Proposed Backport"},"description":"","default_value":0,"optional":true}},"removable_reviewers":[],"reviewers":{"REVIEWER":[{"_account_id":1004,"name":"Mohammed Naser","email":"mnaser@vexxhost.com","username":"mnaser"},{"_account_id":6547,"name":"Andreas Jaeger","email":"jaegerandi@gmail.com","username":"jaegerandi"},{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},{"_account_id":28619,"name":"Dmitriy Rabotyagov","email":"noonedeadpunk@gmail.com","username":"noonedeadpunk"}]},"pending_reviewers":{},"reviewer_updates":[{"updated":"2019-08-27 12:35:39.000000000","updated_by":{"_account_id":28619,"name":"Dmitriy Rabotyagov","email":"noonedeadpunk@gmail.com","username":"noonedeadpunk"},"reviewer":{"_account_id":28619,"name":"Dmitriy Rabotyagov","email":"noonedeadpunk@gmail.com","username":"noonedeadpunk"},"state":"REVIEWER"},{"updated":"2019-08-27 13:01:39.000000000","updated_by":{"_account_id":1004,"name":"Mohammed Naser","email":"mnaser@vexxhost.com","username":"mnaser"},"reviewer":{"_account_id":1004,"name":"Mohammed Naser","email":"mnaser@vexxhost.com","username":"mnaser"},"state":"REVIEWER"},{"updated":"2019-08-27 13:56:58.000000000","updated_by":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"reviewer":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"state":"REVIEWER"}],"messages":[{"id":"61a43874a67cdaec27f6fcef06d5c30cf638fcb7","author":{"_account_id":6547,"name":"Andreas Jaeger","email":"jaegerandi@gmail.com","username":"jaegerandi"},"date":"2019-08-19 07:42:19.000000000","message":"Uploaded patch set 1.","accounts_in_message":[],"_revision_number":1},{"id":"9d9e0f0bec3be8490228ed2abe0c842618bba82e","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2019-08-19 08:37:46.000000000","message":"Patch Set 1: Verified-1\n\nBuild failed (check pipeline).  For information on how to proceed, see\nhttp://docs.openstack.org/infra/manual/developers.html#automated-testing\n\n\n- build-openstack-sphinx-docs https://zuul.opendev.org/t/openstack/build/1adcc672bd1d40f8b286a8a697172eb6 : SUCCESS in 3m 49s\n- openstack-ansible-linters https://zuul.opendev.org/t/openstack/build/c7845809d65f4a098014a1dfd3e7dc6b : SUCCESS in 5m 50s\n- openstack-ansible-functional-centos-7 https://zuul.opendev.org/t/openstack/build/2cc117ff756a423fad0501bd2db21256 : POST_FAILURE in 50m 09s\n- openstack-ansible-functional-ubuntu-xenial https://zuul.opendev.org/t/openstack/build/2ecbc54e25314f1a9a31f14fd51b2f83 : SUCCESS in 27m 54s","accounts_in_message":[],"_revision_number":1},{"id":"a1bc62e40c56aeabb7af362a07f8e1e98318502e","author":{"_account_id":6547,"name":"Andreas Jaeger","email":"jaegerandi@gmail.com","username":"jaegerandi"},"date":"2019-08-20 09:35:11.000000000","message":"Topic changed from goodbye-42.3 to goodbye_42.3","accounts_in_message":[],"_revision_number":1},{"id":"c42f1c98578077da129b6959c7e9d46d6ace0b5b","author":{"_account_id":6547,"name":"Andreas Jaeger","email":"jaegerandi@gmail.com","username":"jaegerandi"},"date":"2019-08-26 20:01:26.000000000","message":"Patch Set 1:\n\nrecheck - fix for POST_FAILURE merged","accounts_in_message":[],"_revision_number":1},{"id":"a02f099551ee17baad17135f9fe84cc1ee7ba8d0","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2019-08-26 20:55:54.000000000","message":"Patch Set 1: Verified+1\n\nBuild succeeded (check pipeline).\n\n- build-openstack-sphinx-docs https://zuul.opendev.org/t/openstack/build/2c55a6aca2a5491fb600055dc4b0c430 : SUCCESS in 4m 15s\n- openstack-ansible-linters https://zuul.opendev.org/t/openstack/build/8875a0fd338b4f929ec87f78ec92bcf7 : SUCCESS in 5m 59s\n- openstack-ansible-functional-centos-7 https://zuul.opendev.org/t/openstack/build/cfbcd006862c4c138d2b0bdc9d31e039 : SUCCESS in 49m 08s\n- openstack-ansible-functional-ubuntu-xenial https://zuul.opendev.org/t/openstack/build/296b62e74b94491aa4cb40fc0ef61df1 : SUCCESS in 29m 54s","accounts_in_message":[],"_revision_number":1},{"id":"e93faead23df2c809ced5168e5befaa72b68f748","author":{"_account_id":28619,"name":"Dmitriy Rabotyagov","email":"noonedeadpunk@gmail.com","username":"noonedeadpunk"},"date":"2019-08-27 12:35:39.000000000","message":"Patch Set 1: Code-Review+2","accounts_in_message":[],"_revision_number":1},{"id":"24a1e5efcb0aa4cf5a08d1be1db21fd82bd48aa5","author":{"_account_id":1004,"name":"Mohammed Naser","email":"mnaser@vexxhost.com","username":"mnaser"},"date":"2019-08-27 13:01:39.000000000","message":"Patch Set 1: Code-Review+2 Workflow+1","accounts_in_message":[],"_revision_number":1},{"id":"0cbd647bdd7409655a222f0921b8cc868644128e","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2019-08-27 13:01:53.000000000","message":"Patch Set 1: -Verified\n\nStarting gate jobs.","accounts_in_message":[],"_revision_number":1},{"id":"12eb986eb78842896d809f32b45391e5d596f94c","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2019-08-27 13:56:58.000000000","message":"Patch Set 1: Verified+2\n\nBuild succeeded (gate pipeline).\n\n- build-openstack-sphinx-docs https://zuul.opendev.org/t/openstack/build/82e9cc0c501f45ef889a1d69c872fa8f : SUCCESS in 3m 43s\n- openstack-ansible-linters https://zuul.opendev.org/t/openstack/build/df6c687d287e4a6a970fa7f39dd86fd2 : SUCCESS in 7m 56s\n- openstack-ansible-functional-centos-7 https://zuul.opendev.org/t/openstack/build/96f069e112b24f2cbef2b08b897e7b25 : SUCCESS in 50m 48s\n- openstack-ansible-functional-ubuntu-xenial https://zuul.opendev.org/t/openstack/build/1989d2eba42645ddbcc7a37f6906a18a : SUCCESS in 30m 44s","accounts_in_message":[],"_revision_number":1},{"id":"dafa5cb7465db2750a2b07c65daf68bbd15af6bf","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2019-08-27 13:56:59.000000000","message":"Change has been successfully merged by Zuul","accounts_in_message":[],"_revision_number":1}],"current_revision_number":1,"current_revision":"fae312554e964c570b79097807cb282da2273c23","revisions":{"fae312554e964c570b79097807cb282da2273c23":{"kind":"REWORK","_number":1,"created":"2019-08-19 07:42:19.000000000","uploader":{"_account_id":6547,"name":"Andreas Jaeger","email":"jaegerandi@gmail.com","username":"jaegerandi"},"ref":"refs/changes/30/677130/1","fetch":{"anonymous http":{"url":"https://review.opendev.org/openstack/openstack-ansible-os_aodh","ref":"refs/changes/30/677130/1","commands":{"Checkout":"git fetch https://review.opendev.org/openstack/openstack-ansible-os_aodh refs/changes/30/677130/1 \u0026\u0026 git checkout FETCH_HEAD","Cherry Pick":"git fetch https://review.opendev.org/openstack/openstack-ansible-os_aodh refs/changes/30/677130/1 \u0026\u0026 git cherry-pick FETCH_HEAD","Format Patch":"git fetch https://review.opendev.org/openstack/openstack-ansible-os_aodh refs/changes/30/677130/1 \u0026\u0026 git format-patch -1 --stdout FETCH_HEAD","Pull":"git pull https://review.opendev.org/openstack/openstack-ansible-os_aodh refs/changes/30/677130/1"}}},"commit":{"parents":[{"commit":"704e177e50af0617d5170eb84f40036a3af4a125","subject":"Updated from OpenStack Ansible Tests","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/openstack-ansible-os_aodh/commit/704e177e50af0617d5170eb84f40036a3af4a125"}]}],"author":{"name":"Andreas Jaeger","email":"aj@suse.com","date":"2019-08-19 07:41:43.000000000","tz":120},"committer":{"name":"Andreas Jaeger","email":"aj@suse.com","date":"2019-08-19 07:41:43.000000000","tz":120},"subject":"Remove openSUSE 42.3 job","message":"Remove openSUSE 42.3 job\n\nopenSUSE 42.3 is dead, remove the job so that the nodes can be removed\nfrom Zuul.\n\nChange-Id: I9d28aecb323ffac57bb062dea46f27fc545a2b02\n","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/openstack-ansible-os_aodh/commit/fae312554e964c570b79097807cb282da2273c23"}],"resolve_conflicts_web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/openstack-ansible-os_aodh/commit/fae312554e964c570b79097807cb282da2273c23"}]},"branch":"refs/heads/stable/pike"}},"requirements":[],"submit_records":[],"submit_requirements":[]}
