)]}'
{"id":"openstack%2Fopenstack-ansible-os_aodh~677438","triplet_id":"openstack%2Fopenstack-ansible-os_aodh~stable%2Frocky~Ie5a28bea3dee410dc7a05b1fc0bccf5b69c9415f","project":"openstack/openstack-ansible-os_aodh","branch":"stable/rocky","topic":"goodbye_42.3","hashtags":[],"change_id":"Ie5a28bea3dee410dc7a05b1fc0bccf5b69c9415f","subject":"Drop openSUSE 42.3 job","status":"MERGED","created":"2019-08-20 09:36:08.000000000","updated":"2019-08-27 13:58:47.000000000","submitted":"2019-08-27 13:58:47.000000000","submitter":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"total_comment_count":0,"unresolved_comment_count":0,"has_review_started":true,"submission_id":"677438-1566914327425-22bc38ef","meta_rev_id":"06312ab2a315a35b9102e529cdc7caa87c396eb8","_number":677438,"virtual_id_number":677438,"owner":{"_account_id":6547,"name":"Andreas Jaeger","email":"jaegerandi@gmail.com","username":"jaegerandi"},"actions":{},"labels":{"Verified":{"approved":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"all":[{"value":0,"date":"2019-08-26 20:01:40.000000000","_account_id":6547,"name":"Andreas Jaeger","email":"jaegerandi@gmail.com","username":"jaegerandi"},{"value":0,"_account_id":28619,"name":"Dmitriy Rabotyagov","email":"noonedeadpunk@gmail.com","username":"noonedeadpunk"},{"value":2,"date":"2019-08-27 13:58:47.000000000","post_submit":true,"permitted_voting_range":{"min":2,"max":2},"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},{"value":0,"_account_id":1004,"name":"Mohammed Naser","email":"mnaser@vexxhost.com","username":"mnaser"}],"values":{"-2":"Fails","-1":"Doesn\u0027t seem to work"," 0":"No score","+1":"Works for me","+2":"Verified"},"description":"","default_value":0,"optional":true},"Code-Review":{"approved":{"_account_id":28619,"name":"Dmitriy Rabotyagov","email":"noonedeadpunk@gmail.com","username":"noonedeadpunk"},"all":[{"value":0,"_account_id":6547,"name":"Andreas Jaeger","email":"jaegerandi@gmail.com","username":"jaegerandi"},{"value":2,"date":"2019-08-27 12:35:29.000000000","permitted_voting_range":{"min":2,"max":2},"_account_id":28619,"name":"Dmitriy Rabotyagov","email":"noonedeadpunk@gmail.com","username":"noonedeadpunk"},{"value":0,"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},{"value":2,"date":"2019-08-27 13:01:42.000000000","permitted_voting_range":{"min":2,"max":2},"_account_id":1004,"name":"Mohammed Naser","email":"mnaser@vexxhost.com","username":"mnaser"}],"values":{"-2":"Do not merge","-1":"This patch needs further work before it can be merged"," 0":"No score","+1":"Looks good to me, but someone else must approve","+2":"Looks good to me (core reviewer)"},"description":"","default_value":0,"optional":true},"Workflow":{"approved":{"_account_id":1004,"name":"Mohammed Naser","email":"mnaser@vexxhost.com","username":"mnaser"},"all":[{"value":0,"_account_id":6547,"name":"Andreas Jaeger","email":"jaegerandi@gmail.com","username":"jaegerandi"},{"value":0,"_account_id":28619,"name":"Dmitriy Rabotyagov","email":"noonedeadpunk@gmail.com","username":"noonedeadpunk"},{"value":0,"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},{"value":1,"date":"2019-08-27 13:01:42.000000000","permitted_voting_range":{"min":1,"max":1},"_account_id":1004,"name":"Mohammed Naser","email":"mnaser@vexxhost.com","username":"mnaser"}],"values":{"-1":"Work in progress"," 0":"Ready for reviews","+1":"Approved"},"description":"","default_value":0,"optional":true},"Backport-Candidate":{"all":[{"value":0,"_account_id":6547,"name":"Andreas Jaeger","email":"jaegerandi@gmail.com","username":"jaegerandi"},{"value":0,"_account_id":28619,"name":"Dmitriy Rabotyagov","email":"noonedeadpunk@gmail.com","username":"noonedeadpunk"},{"value":0,"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},{"value":0,"_account_id":1004,"name":"Mohammed Naser","email":"mnaser@vexxhost.com","username":"mnaser"}],"values":{"-1":"Not A Backport Candidate"," 0":"Backport Review Needed","+1":"Proposed Backport"},"description":"","default_value":0,"optional":true}},"removable_reviewers":[],"reviewers":{"REVIEWER":[{"_account_id":1004,"name":"Mohammed Naser","email":"mnaser@vexxhost.com","username":"mnaser"},{"_account_id":6547,"name":"Andreas Jaeger","email":"jaegerandi@gmail.com","username":"jaegerandi"},{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},{"_account_id":28619,"name":"Dmitriy Rabotyagov","email":"noonedeadpunk@gmail.com","username":"noonedeadpunk"}]},"pending_reviewers":{},"reviewer_updates":[{"updated":"2019-08-27 12:35:29.000000000","updated_by":{"_account_id":28619,"name":"Dmitriy Rabotyagov","email":"noonedeadpunk@gmail.com","username":"noonedeadpunk"},"reviewer":{"_account_id":28619,"name":"Dmitriy Rabotyagov","email":"noonedeadpunk@gmail.com","username":"noonedeadpunk"},"state":"REVIEWER"},{"updated":"2019-08-27 13:01:42.000000000","updated_by":{"_account_id":1004,"name":"Mohammed Naser","email":"mnaser@vexxhost.com","username":"mnaser"},"reviewer":{"_account_id":1004,"name":"Mohammed Naser","email":"mnaser@vexxhost.com","username":"mnaser"},"state":"REVIEWER"},{"updated":"2019-08-27 13:58:47.000000000","updated_by":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"reviewer":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"state":"REVIEWER"}],"messages":[{"id":"d5fc9309cdf352e91bdae76eb1509cd4f30e1c58","author":{"_account_id":6547,"name":"Andreas Jaeger","email":"jaegerandi@gmail.com","username":"jaegerandi"},"date":"2019-08-20 09:36:08.000000000","message":"Uploaded patch set 1.","accounts_in_message":[],"_revision_number":1},{"id":"50eb6e752619591270a54613794d54fe280bc4b1","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2019-08-20 10:22:20.000000000","message":"Patch Set 1: Verified-1\n\nBuild failed (check pipeline).  For information on how to proceed, see\nhttp://docs.openstack.org/infra/manual/developers.html#automated-testing\n\n\n- openstack-ansible-linters https://zuul.opendev.org/t/openstack/build/493c19ec2d00472887b1886d7423f10f : SUCCESS in 6m 39s\n- openstack-ansible-functional-centos-7 https://zuul.opendev.org/t/openstack/build/5d6565ead8a64238a6d2250c247485a6 : POST_FAILURE in 43m 26s\n- openstack-ansible-functional-opensuse-423 https://zuul.opendev.org/t/openstack/build/fb5fc1ddb5484ebea15fa0cba00bb731 : POST_FAILURE in 6m 47s (non-voting)\n- openstack-ansible-functional-opensuse-150 https://zuul.opendev.org/t/openstack/build/f2ae7b7c0bc7441f848fcf3823802a59 : POST_FAILURE in 11m 15s (non-voting)\n- openstack-ansible-functional-ubuntu-bionic https://zuul.opendev.org/t/openstack/build/e94d6a203480444abc37325f92441f32 : POST_FAILURE in 34m 41s\n- openstack-ansible-functional-ubuntu-xenial https://zuul.opendev.org/t/openstack/build/c0b4b1159fb741f3aa1f58942e78a322 : POST_FAILURE in 35m 51s\n- build-openstack-sphinx-docs https://zuul.opendev.org/t/openstack/build/bed0901cc78c4d318aabb60f9d30f9ea : SUCCESS in 3m 38s\n- openstack-ansible-functional-distro_install-ubuntu-xenial https://zuul.opendev.org/t/openstack/build/ef046ca546f144aa8c3b56bf4a02b159 : POST_FAILURE in 38m 12s\n- openstack-ansible-functional-distro_install-centos-7 https://zuul.opendev.org/t/openstack/build/91a5a420678841f392d3b096076efe4a : POST_FAILURE in 42m 25s","accounts_in_message":[],"_revision_number":1},{"id":"85d12541316b178e88dea3209c758dac01cccd2c","author":{"_account_id":17068,"name":"Jean-Philippe Evrard","email":"openstack@a.spamming.party","username":"evrardjp"},"date":"2019-08-22 10:37:57.000000000","message":"Patch Set 2: Commit message was updated.","accounts_in_message":[],"_revision_number":2},{"id":"ff49720ad43c096700339fbdb5880333cd6be036","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2019-08-22 11:26:47.000000000","message":"Patch Set 2: Verified-1\n\nBuild failed (check pipeline).  For information on how to proceed, see\nhttp://docs.openstack.org/infra/manual/developers.html#automated-testing\n\n\n- openstack-ansible-linters https://zuul.opendev.org/t/openstack/build/724b82bbe1d84c00a94bf507613b2bf9 : SUCCESS in 7m 18s\n- openstack-ansible-functional-centos-7 https://zuul.opendev.org/t/openstack/build/1f7a2c77a6c242059fb6cdc3a3ccc09f : POST_FAILURE in 40m 34s\n- openstack-ansible-functional-opensuse-150 https://zuul.opendev.org/t/openstack/build/077d1e7dfdcf4d6683ac8734347e7aaf : POST_FAILURE in 14m 32s (non-voting)\n- openstack-ansible-functional-ubuntu-bionic https://zuul.opendev.org/t/openstack/build/047f3d7b7b4c46dfb8379c4f99fc5b9b : POST_FAILURE in 33m 52s\n- openstack-ansible-functional-ubuntu-xenial https://zuul.opendev.org/t/openstack/build/6d4a23500f344564a15a8703728fc689 : POST_FAILURE in 37m 24s\n- build-openstack-sphinx-docs https://zuul.opendev.org/t/openstack/build/1e3f5e47e4594c3c9fb36c3a3651d074 : SUCCESS in 6m 10s\n- openstack-ansible-functional-distro_install-ubuntu-xenial https://zuul.opendev.org/t/openstack/build/4eea913ff65846b68f5b525f2d694838 : POST_FAILURE in 39m 51s\n- openstack-ansible-functional-distro_install-centos-7 https://zuul.opendev.org/t/openstack/build/bcda976bc6294706a40c039a15b10a55 : POST_FAILURE in 36m 31s","accounts_in_message":[],"_revision_number":2},{"id":"15c5dcb502e309a30532e492fe9af449f2608435","author":{"_account_id":6547,"name":"Andreas Jaeger","email":"jaegerandi@gmail.com","username":"jaegerandi"},"date":"2019-08-26 20:01:40.000000000","message":"Patch Set 2:\n\nrecheck - fix for POST_FAILURE merged","accounts_in_message":[],"_revision_number":2},{"id":"726c1b755d5f8a5ebfcd148dc8381781c7d9086f","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2019-08-26 20:56:29.000000000","message":"Patch Set 2: Verified+1\n\nBuild succeeded (check pipeline).\n\n- openstack-ansible-linters https://zuul.opendev.org/t/openstack/build/82a06b85828e47b99c0585ddb5d24977 : SUCCESS in 7m 02s\n- openstack-ansible-functional-centos-7 https://zuul.opendev.org/t/openstack/build/0a1e0e4694ec4b4592b9aa018c22d4ec : SUCCESS in 48m 37s\n- openstack-ansible-functional-opensuse-150 https://zuul.opendev.org/t/openstack/build/64950a139d674d48b1747615e1ab0b74 : RETRY_LIMIT in 4m 49s (non-voting)\n- openstack-ansible-functional-ubuntu-bionic https://zuul.opendev.org/t/openstack/build/66f69254c0084669a6e81f517cc8cec6 : SUCCESS in 30m 24s\n- openstack-ansible-functional-ubuntu-xenial https://zuul.opendev.org/t/openstack/build/21fdd5ed87a64a45ad88e0477bff7a94 : SUCCESS in 37m 45s\n- build-openstack-sphinx-docs https://zuul.opendev.org/t/openstack/build/29231f771ccf4e38b3b6d8e87a837c87 : SUCCESS in 5m 02s\n- openstack-ansible-functional-distro_install-ubuntu-xenial https://zuul.opendev.org/t/openstack/build/01614239f16a434bad862ef212259fbe : SUCCESS in 33m 43s\n- openstack-ansible-functional-distro_install-centos-7 https://zuul.opendev.org/t/openstack/build/65f79b48acb447de95225e0c25926179 : SUCCESS in 36m 23s","accounts_in_message":[],"_revision_number":2},{"id":"c5ec59b571405d0232217b07adab406bef971da0","author":{"_account_id":28619,"name":"Dmitriy Rabotyagov","email":"noonedeadpunk@gmail.com","username":"noonedeadpunk"},"date":"2019-08-27 12:35:29.000000000","message":"Patch Set 2: Code-Review+2","accounts_in_message":[],"_revision_number":2},{"id":"52bb2104d6c01491b65895623653654e5f8cca0a","author":{"_account_id":1004,"name":"Mohammed Naser","email":"mnaser@vexxhost.com","username":"mnaser"},"date":"2019-08-27 13:01:42.000000000","message":"Patch Set 2: Code-Review+2 Workflow+1","accounts_in_message":[],"_revision_number":2},{"id":"32f869273f07749c0da9b237375a027f3c79c677","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2019-08-27 13:01:57.000000000","message":"Patch Set 2: -Verified\n\nStarting gate jobs.","accounts_in_message":[],"_revision_number":2},{"id":"9475fa9dec5a59f7b34afa5ad3ca1121a5e12aa9","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2019-08-27 13:58:47.000000000","message":"Change has been successfully merged by Zuul","accounts_in_message":[],"_revision_number":2},{"id":"292b272ebed9ecff296a3968ecc424e20c5d40cd","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2019-08-27 13:58:47.000000000","message":"Patch Set 2: Verified+2\n\nBuild succeeded (gate pipeline).\n\n- openstack-ansible-linters https://zuul.opendev.org/t/openstack/build/736dd1525b9e474e8b052e4db365fc41 : SUCCESS in 9m 50s\n- openstack-ansible-functional-centos-7 https://zuul.opendev.org/t/openstack/build/92d8cd8f2a264c7ea870abef39ef3ac8 : SUCCESS in 41m 13s\n- openstack-ansible-functional-ubuntu-bionic https://zuul.opendev.org/t/openstack/build/0e3424dc3ba341cfb4518a24cdf3aafb : SUCCESS in 34m 22s\n- openstack-ansible-functional-ubuntu-xenial https://zuul.opendev.org/t/openstack/build/a360eeee601e49d5b635af04c5f3ad6e : SUCCESS in 37m 43s\n- build-openstack-sphinx-docs https://zuul.opendev.org/t/openstack/build/da55a015ef984fd2aa198b781a05cd60 : SUCCESS in 5m 34s\n- openstack-ansible-functional-distro_install-ubuntu-xenial https://zuul.opendev.org/t/openstack/build/eeac42158bcd416c846f99aa3d073148 : SUCCESS in 45m 41s\n- openstack-ansible-functional-distro_install-centos-7 https://zuul.opendev.org/t/openstack/build/c4f689d5d928467cbd1bfb83d803a9a9 : SUCCESS in 41m 01s","accounts_in_message":[],"_revision_number":2}],"current_revision_number":2,"current_revision":"6ccf1ab9ea1a7bb8588a899ed2defbb4d8b4c5ae","revisions":{"fdf14936f9006371c7dee05ebd040fb3454b2826":{"kind":"REWORK","_number":1,"created":"2019-08-20 09:36:08.000000000","uploader":{"_account_id":6547,"name":"Andreas Jaeger","email":"jaegerandi@gmail.com","username":"jaegerandi"},"ref":"refs/changes/38/677438/1","fetch":{"anonymous http":{"url":"https://review.opendev.org/openstack/openstack-ansible-os_aodh","ref":"refs/changes/38/677438/1","commands":{"Checkout":"git fetch https://review.opendev.org/openstack/openstack-ansible-os_aodh refs/changes/38/677438/1 \u0026\u0026 git checkout FETCH_HEAD","Cherry Pick":"git fetch https://review.opendev.org/openstack/openstack-ansible-os_aodh refs/changes/38/677438/1 \u0026\u0026 git cherry-pick FETCH_HEAD","Format Patch":"git fetch https://review.opendev.org/openstack/openstack-ansible-os_aodh refs/changes/38/677438/1 \u0026\u0026 git format-patch -1 --stdout FETCH_HEAD","Pull":"git pull https://review.opendev.org/openstack/openstack-ansible-os_aodh refs/changes/38/677438/1"}}},"commit":{"parents":[{"commit":"fde216d7e4c39adc7fa955ad8bd3a42bb0625dfa","subject":"Updated from OpenStack Ansible Tests","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/openstack-ansible-os_aodh/commit/fde216d7e4c39adc7fa955ad8bd3a42bb0625dfa"}]}],"author":{"name":"Andreas Jaeger","email":"aj@suse.com","date":"2019-08-20 09:34:30.000000000","tz":120},"committer":{"name":"Andreas Jaeger","email":"aj@suse.com","date":"2019-08-20 09:34:30.000000000","tz":120},"subject":"Drop openSUSE 42.3 job","message":"Drop openSUSE 42.3 job\n\n42.3 will not be maintained anymore and therefore\nthere is no reason to test it. Users are advised\nto upgrade to Leap 15, and we should do the same.\n\nRemove openSUSE 42.3 so that we can remove it from CI.\n\nChange-Id: Ie5a28bea3dee410dc7a05b1fc0bccf5b69c9415f\n","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/openstack-ansible-os_aodh/commit/fdf14936f9006371c7dee05ebd040fb3454b2826"}],"resolve_conflicts_web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/openstack-ansible-os_aodh/commit/fdf14936f9006371c7dee05ebd040fb3454b2826"}]},"branch":"refs/heads/stable/rocky"},"6ccf1ab9ea1a7bb8588a899ed2defbb4d8b4c5ae":{"kind":"NO_CODE_CHANGE","_number":2,"created":"2019-08-22 10:37:57.000000000","uploader":{"_account_id":17068,"name":"Jean-Philippe Evrard","email":"openstack@a.spamming.party","username":"evrardjp"},"ref":"refs/changes/38/677438/2","fetch":{"anonymous http":{"url":"https://review.opendev.org/openstack/openstack-ansible-os_aodh","ref":"refs/changes/38/677438/2","commands":{"Checkout":"git fetch https://review.opendev.org/openstack/openstack-ansible-os_aodh refs/changes/38/677438/2 \u0026\u0026 git checkout FETCH_HEAD","Cherry Pick":"git fetch https://review.opendev.org/openstack/openstack-ansible-os_aodh refs/changes/38/677438/2 \u0026\u0026 git cherry-pick FETCH_HEAD","Format Patch":"git fetch https://review.opendev.org/openstack/openstack-ansible-os_aodh refs/changes/38/677438/2 \u0026\u0026 git format-patch -1 --stdout FETCH_HEAD","Pull":"git pull https://review.opendev.org/openstack/openstack-ansible-os_aodh refs/changes/38/677438/2"}}},"commit":{"parents":[{"commit":"fde216d7e4c39adc7fa955ad8bd3a42bb0625dfa","subject":"Updated from OpenStack Ansible Tests","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/openstack-ansible-os_aodh/commit/fde216d7e4c39adc7fa955ad8bd3a42bb0625dfa"}]}],"author":{"name":"Andreas Jaeger","email":"aj@suse.com","date":"2019-08-20 09:34:30.000000000","tz":120},"committer":{"name":"Jean-Philippe Evrard","email":"jean-philippe@evrard.me","date":"2019-08-22 10:37:55.000000000","tz":0},"subject":"Drop openSUSE 42.3 job","message":"Drop openSUSE 42.3 job\n\n42.3 will not be maintained anymore and therefore\nthere is no reason to test it. Users are advised\nto upgrade to Leap 15, and we should do the same.\n\nRemove openSUSE 42.3 so that we can remove it from CI.\n\nDepends-On: https://review.opendev.org/677947\nChange-Id: Ie5a28bea3dee410dc7a05b1fc0bccf5b69c9415f\n","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/openstack-ansible-os_aodh/commit/6ccf1ab9ea1a7bb8588a899ed2defbb4d8b4c5ae"}],"resolve_conflicts_web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/openstack-ansible-os_aodh/commit/6ccf1ab9ea1a7bb8588a899ed2defbb4d8b4c5ae"}]},"branch":"refs/heads/stable/rocky"}},"requirements":[],"submit_records":[],"submit_requirements":[]}
