)]}'
{"id":"openstack%2Fopenstack-ansible-os_aodh~680035","triplet_id":"openstack%2Fopenstack-ansible-os_aodh~master~Ie6ed1378f23645e71a80019f33b9ce482bb43680","project":"openstack/openstack-ansible-os_aodh","branch":"master","topic":"openstack/openstack-ansible-tests/sync-tests","hashtags":[],"change_id":"Ie6ed1378f23645e71a80019f33b9ce482bb43680","subject":"Updated from OpenStack Ansible Tests","status":"MERGED","created":"2019-09-04 15:11:48.000000000","updated":"2019-09-06 22:54:06.000000000","submitted":"2019-09-06 22:52:25.000000000","submitter":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"total_comment_count":0,"unresolved_comment_count":0,"has_review_started":true,"submission_id":"680035-1567810345336-707fe744","meta_rev_id":"078c80825c77f2bb713fe7a2c07068700fc834a8","_number":680035,"virtual_id_number":680035,"owner":{"_account_id":11131,"name":"OpenStack Proposal Bot","username":"proposal-bot","tags":["SERVICE_USER"]},"actions":{},"labels":{"Verified":{"approved":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"all":[{"value":0,"_account_id":28619,"name":"Dmitriy Rabotyagov","email":"noonedeadpunk@gmail.com","username":"noonedeadpunk"},{"value":2,"date":"2019-09-06 22:52:25.000000000","post_submit":true,"permitted_voting_range":{"min":2,"max":2},"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]}],"values":{"-2":"Fails","-1":"Doesn\u0027t seem to work"," 0":"No score","+1":"Works for me","+2":"Verified"},"description":"","default_value":0,"optional":true},"Code-Review":{"approved":{"_account_id":28619,"name":"Dmitriy Rabotyagov","email":"noonedeadpunk@gmail.com","username":"noonedeadpunk"},"all":[{"value":2,"date":"2019-09-06 20:38:47.000000000","permitted_voting_range":{"min":2,"max":2},"_account_id":28619,"name":"Dmitriy Rabotyagov","email":"noonedeadpunk@gmail.com","username":"noonedeadpunk"},{"value":0,"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]}],"values":{"-2":"Do not merge","-1":"This patch needs further work before it can be merged"," 0":"No score","+1":"Looks good to me, but someone else must approve","+2":"Looks good to me (core reviewer)"},"description":"","default_value":0,"optional":true},"Workflow":{"approved":{"_account_id":28619,"name":"Dmitriy Rabotyagov","email":"noonedeadpunk@gmail.com","username":"noonedeadpunk"},"all":[{"value":1,"date":"2019-09-06 20:38:47.000000000","permitted_voting_range":{"min":1,"max":1},"_account_id":28619,"name":"Dmitriy Rabotyagov","email":"noonedeadpunk@gmail.com","username":"noonedeadpunk"},{"value":0,"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]}],"values":{"-1":"Work in progress"," 0":"Ready for reviews","+1":"Approved"},"description":"","default_value":0,"optional":true},"Backport-Candidate":{"all":[{"value":0,"_account_id":28619,"name":"Dmitriy Rabotyagov","email":"noonedeadpunk@gmail.com","username":"noonedeadpunk"},{"value":0,"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]}],"values":{"-1":"Not A Backport Candidate"," 0":"Backport Review Needed","+1":"Proposed Backport"},"description":"","default_value":0,"optional":true}},"removable_reviewers":[],"reviewers":{"REVIEWER":[{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},{"_account_id":28619,"name":"Dmitriy Rabotyagov","email":"noonedeadpunk@gmail.com","username":"noonedeadpunk"}]},"pending_reviewers":{},"reviewer_updates":[{"updated":"2019-09-06 20:38:47.000000000","updated_by":{"_account_id":28619,"name":"Dmitriy Rabotyagov","email":"noonedeadpunk@gmail.com","username":"noonedeadpunk"},"reviewer":{"_account_id":28619,"name":"Dmitriy Rabotyagov","email":"noonedeadpunk@gmail.com","username":"noonedeadpunk"},"state":"REVIEWER"},{"updated":"2019-09-06 22:52:25.000000000","updated_by":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"reviewer":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"state":"REVIEWER"}],"messages":[{"id":"83c0138edee65b247c4630b5c21b55077c79fd26","author":{"_account_id":11131,"name":"OpenStack Proposal Bot","username":"proposal-bot","tags":["SERVICE_USER"]},"date":"2019-09-04 15:11:48.000000000","message":"Uploaded patch set 1.","accounts_in_message":[],"_revision_number":1},{"id":"c2229a3c9136d6f09ce7da4c1ffe0ed3ff161605","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2019-09-04 18:33:46.000000000","message":"Patch Set 1: Verified-1\n\nBuild failed (check pipeline).  For information on how to proceed, see\nhttp://docs.openstack.org/infra/manual/developers.html#automated-testing\n\n\n- openstack-ansible-linters https://zuul.opendev.org/t/openstack/build/fa87f8cb1a0b4a178ca7cf80b8ce0cc6 : SUCCESS in 10m 40s\n- openstack-ansible-deploy-aio_metal-debian-stable https://zuul.opendev.org/t/openstack/build/b877dc829567494eaf832bdc0d6315d9 : POST_FAILURE in 1h 15m 10s\n- openstack-ansible-deploy-aio_metal-centos-7 https://zuul.opendev.org/t/openstack/build/216363da2797450d885674c746b03081 : SUCCESS in 2h 07m 00s\n- openstack-ansible-deploy-aio_metal-opensuse-15 https://zuul.opendev.org/t/openstack/build/1db9395f52284d22820f839fdc286e27 : RETRY_LIMIT in 4m 24s\n- openstack-ansible-deploy-aio_metal-ubuntu-bionic https://zuul.opendev.org/t/openstack/build/a9108800485e43f8bb5bc54c36e13fcb : POST_FAILURE in 1h 19m 27s\n- openstack-ansible-deploy-aio_distro_metal-centos-7 https://zuul.opendev.org/t/openstack/build/e3ef8553a8f64afc856064f4b6ec492d : SUCCESS in 1h 31m 08s\n- openstack-ansible-deploy-aio_distro_metal-opensuse-15 https://zuul.opendev.org/t/openstack/build/707c8d3b3d2242fa884b79bb2d6d705f : FAILURE in 31m 05s (non-voting)\n- openstack-ansible-deploy-aio_distro_metal-ubuntu-bionic https://zuul.opendev.org/t/openstack/build/88048947f7a444308b63a5a0cd04ae4c : POST_FAILURE in 58m 48s\n- openstack-ansible-deploy-aio_telemetry_metal-centos-7 https://zuul.opendev.org/t/openstack/build/a644efe0cf664c7d86ffbcb32d72fc45 : POST_FAILURE in 1h 39m 32s\n- openstack-ansible-deploy-aio_telemetry_metal-debian-stable https://zuul.opendev.org/t/openstack/build/6c8438bf4ba1498681cb23603b04be8c : SUCCESS in 1h 27m 44s\n- openstack-ansible-deploy-aio_telemetry_metal-ubuntu-bionic https://zuul.opendev.org/t/openstack/build/d5da94e6eec94eb1a96c292f80feac46 : SUCCESS in 1h 38m 46s\n- openstack-tox-docs https://zuul.opendev.org/t/openstack/build/b4b97dca1f684ed0a67ddbcb5ca9b26e : SUCCESS in 5m 14s","accounts_in_message":[],"_revision_number":1},{"id":"f6576b1a141861be74707e92f2ebd527b2876d17","author":{"_account_id":28619,"name":"Dmitriy Rabotyagov","email":"noonedeadpunk@gmail.com","username":"noonedeadpunk"},"date":"2019-09-05 14:35:58.000000000","message":"Patch Set 1:\n\nrecheck","accounts_in_message":[],"_revision_number":1},{"id":"d87bb1305ecd48e09f0f9dd9e06f911c7832b47b","author":{"_account_id":28619,"name":"Dmitriy Rabotyagov","email":"noonedeadpunk@gmail.com","username":"noonedeadpunk"},"date":"2019-09-05 14:46:46.000000000","message":"Patch Set 1:\n\nrecheck","accounts_in_message":[],"_revision_number":1},{"id":"059c9669a1187adbb787044f9649271a482cc7db","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2019-09-05 18:20:01.000000000","message":"Patch Set 1:\n\nBuild failed (check pipeline).  For information on how to proceed, see\nhttp://docs.openstack.org/infra/manual/developers.html#automated-testing\n\n\n- openstack-ansible-linters https://zuul.opendev.org/t/openstack/build/602b58652b7444948da19b3a61b291c1 : SUCCESS in 6m 45s\n- openstack-ansible-deploy-aio_metal-debian-stable https://zuul.opendev.org/t/openstack/build/957c1c3a363948e690ea4d061f97c820 : SUCCESS in 1h 39m 13s\n- openstack-ansible-deploy-aio_metal-centos-7 https://zuul.opendev.org/t/openstack/build/de1e1f1ed0aa416dbadf35b87542d727 : SUCCESS in 1h 56m 34s\n- openstack-ansible-deploy-aio_metal-opensuse-15 https://zuul.opendev.org/t/openstack/build/ba59cdc5fdf14f069ef12b90c38280c5 : SUCCESS in 1h 39m 12s\n- openstack-ansible-deploy-aio_metal-ubuntu-bionic https://zuul.opendev.org/t/openstack/build/dcced45c28ca444fba56cbf36ccfd80e : SUCCESS in 1h 24m 10s\n- openstack-ansible-deploy-aio_distro_metal-centos-7 https://zuul.opendev.org/t/openstack/build/7e3b0eff800f4fd9b2af0e0481023c90 : SUCCESS in 1h 30m 29s\n- openstack-ansible-deploy-aio_distro_metal-opensuse-15 https://zuul.opendev.org/t/openstack/build/f483350d92ae42dabad506f7d0dec6bd : FAILURE in 33m 02s (non-voting)\n- openstack-ansible-deploy-aio_distro_metal-ubuntu-bionic https://zuul.opendev.org/t/openstack/build/f37d3322dbee4be3a6411b35a61a06d9 : SUCCESS in 1h 03m 28s\n- openstack-ansible-deploy-aio_telemetry_metal-centos-7 https://zuul.opendev.org/t/openstack/build/563ebca4a9a040bb938a0d003d96124b : SUCCESS in 2h 02m 22s\n- openstack-ansible-deploy-aio_telemetry_metal-debian-stable https://zuul.opendev.org/t/openstack/build/a6c2985757d74092aba16972f4b13f34 : SUCCESS in 1h 55m 03s\n- openstack-ansible-deploy-aio_telemetry_metal-ubuntu-bionic https://zuul.opendev.org/t/openstack/build/e2cd29cd6dae4bb09dca57db4b93a9c8 : FAILURE in 1h 40m 12s\n- openstack-tox-docs https://zuul.opendev.org/t/openstack/build/c6bda9c1a4b24795ad6784b68de86db2 : SUCCESS in 6m 56s","accounts_in_message":[],"_revision_number":1},{"id":"98c635a5fffd3fe72640c75d0cea602ff45fc32f","author":{"_account_id":28619,"name":"Dmitriy Rabotyagov","email":"noonedeadpunk@gmail.com","username":"noonedeadpunk"},"date":"2019-09-05 21:53:53.000000000","message":"Patch Set 1:\n\nrecheck","accounts_in_message":[],"_revision_number":1},{"id":"7c3636cca797e3f915bf3d3c435e678665883478","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2019-09-06 01:01:34.000000000","message":"Patch Set 1:\n\nBuild failed (check pipeline).  For information on how to proceed, see\nhttp://docs.openstack.org/infra/manual/developers.html#automated-testing\n\n\n- openstack-ansible-linters https://zuul.opendev.org/t/openstack/build/ccd75e76300e45168cee139e8736c904 : TIMED_OUT in 15m 59s\n- openstack-ansible-deploy-aio_metal-debian-stable https://zuul.opendev.org/t/openstack/build/e2a75a20d25a4c7d914f9d5bffdb0800 : SUCCESS in 1h 19m 12s\n- openstack-ansible-deploy-aio_metal-centos-7 https://zuul.opendev.org/t/openstack/build/08b0d77b1a4d4ef29c83f8dce9ccd493 : SUCCESS in 1h 49m 55s\n- openstack-ansible-deploy-aio_metal-opensuse-15 https://zuul.opendev.org/t/openstack/build/cd3ecee418464149aaf618b199be21f5 : FAILURE in 1h 00m 24s\n- openstack-ansible-deploy-aio_metal-ubuntu-bionic https://zuul.opendev.org/t/openstack/build/ac48b43ea20f48269f92d7454753cbf1 : SUCCESS in 1h 21m 17s\n- openstack-ansible-deploy-aio_distro_metal-centos-7 https://zuul.opendev.org/t/openstack/build/1b72c4cd62524528938ada4f28ef3cf3 : SUCCESS in 1h 22m 20s\n- openstack-ansible-deploy-aio_distro_metal-opensuse-15 https://zuul.opendev.org/t/openstack/build/0f338a639ab94a53ab924718b7b56592 : FAILURE in 1h 09m 20s (non-voting)\n- openstack-ansible-deploy-aio_distro_metal-ubuntu-bionic https://zuul.opendev.org/t/openstack/build/558bab12f3a147d7ba27b5a666b1675d : SUCCESS in 1h 09m 44s\n- openstack-ansible-deploy-aio_telemetry_metal-centos-7 https://zuul.opendev.org/t/openstack/build/07423cc7b4674ed589411e467db68c71 : SUCCESS in 2h 11m 08s\n- openstack-ansible-deploy-aio_telemetry_metal-debian-stable https://zuul.opendev.org/t/openstack/build/e34b934d0e5843d9b030ad30b2bfa2dc : SUCCESS in 1h 32m 56s\n- openstack-ansible-deploy-aio_telemetry_metal-ubuntu-bionic https://zuul.opendev.org/t/openstack/build/75ad601f745c4a6689af7ec0240c9ef7 : FAILURE in 1h 18m 55s\n- openstack-tox-docs https://zuul.opendev.org/t/openstack/build/c20e86181e9d43109cc2c90d4105f977 : SUCCESS in 3m 38s","accounts_in_message":[],"_revision_number":1},{"id":"89a6b052f6010a0650d792af206ae693b5420e61","author":{"_account_id":28619,"name":"Dmitriy Rabotyagov","email":"noonedeadpunk@gmail.com","username":"noonedeadpunk"},"date":"2019-09-06 07:09:52.000000000","message":"Patch Set 1:\n\nrecheck","accounts_in_message":[],"_revision_number":1},{"id":"479e0017ddfbe8eee1cd5036628d90a9e7dcf194","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2019-09-06 09:31:05.000000000","message":"Patch Set 1:\n\nBuild failed (check pipeline).  For information on how to proceed, see\nhttp://docs.openstack.org/infra/manual/developers.html#automated-testing\n\n\n- openstack-ansible-linters https://zuul.opendev.org/t/openstack/build/fcfb78d2ffea40c0baa8546291f9d627 : SUCCESS in 5m 43s\n- openstack-ansible-deploy-aio_metal-debian-stable https://zuul.opendev.org/t/openstack/build/d38e77e151ae41168bade706b6373758 : SUCCESS in 1h 07m 54s\n- openstack-ansible-deploy-aio_metal-centos-7 https://zuul.opendev.org/t/openstack/build/b9f9e22799ad423abff5dd7ab35aa3df : SUCCESS in 1h 45m 06s\n- openstack-ansible-deploy-aio_metal-opensuse-15 https://zuul.opendev.org/t/openstack/build/8d478174a76343ee8fc7a5344226d30e : FAILURE in 1h 30m 06s\n- openstack-ansible-deploy-aio_metal-ubuntu-bionic https://zuul.opendev.org/t/openstack/build/0d1dfcd268e84ecaa2c4863352ed537f : SUCCESS in 1h 15m 59s\n- openstack-ansible-deploy-aio_distro_metal-centos-7 https://zuul.opendev.org/t/openstack/build/1ac2a99fb0104c069bc24c54c09fae12 : SUCCESS in 1h 18m 05s\n- openstack-ansible-deploy-aio_distro_metal-opensuse-15 https://zuul.opendev.org/t/openstack/build/7ea55c8b0f2f42a5a4a424db2074a2f9 : FAILURE in 28m 54s (non-voting)\n- openstack-ansible-deploy-aio_distro_metal-ubuntu-bionic https://zuul.opendev.org/t/openstack/build/3291f2e871f347db88babd788a083e8d : SUCCESS in 52m 07s\n- openstack-ansible-deploy-aio_telemetry_metal-centos-7 https://zuul.opendev.org/t/openstack/build/aaa542f858eb4542b01ee0ef69fe5152 : SUCCESS in 1h 41m 03s\n- openstack-ansible-deploy-aio_telemetry_metal-debian-stable https://zuul.opendev.org/t/openstack/build/14061a46d11243e5a7a351e0c600ed6e : SUCCESS in 1h 10m 36s\n- openstack-ansible-deploy-aio_telemetry_metal-ubuntu-bionic https://zuul.opendev.org/t/openstack/build/612f857d93ce4c5db8df0bba86b51b4d : SUCCESS in 1h 33m 55s\n- openstack-tox-docs https://zuul.opendev.org/t/openstack/build/a09ff182a24043c68f4a1044744729df : SUCCESS in 2m 59s","accounts_in_message":[],"_revision_number":1},{"id":"f6d9a9aef978fb1846edd5f182c2c1dfdb7f6eb9","author":{"_account_id":28619,"name":"Dmitriy Rabotyagov","email":"noonedeadpunk@gmail.com","username":"noonedeadpunk"},"date":"2019-09-06 11:17:15.000000000","message":"Patch Set 1:\n\nrecheck","accounts_in_message":[],"_revision_number":1},{"id":"50be0fdf3ced44656b26229190199452905eba50","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2019-09-06 13:06:29.000000000","message":"Patch Set 1:\n\nBuild failed (check pipeline).  For information on how to proceed, see\nhttp://docs.openstack.org/infra/manual/developers.html#automated-testing\n\n\n- openstack-ansible-linters https://zuul.opendev.org/t/openstack/build/40819e47800c4351b8d61f8ed1c0a734 : SUCCESS in 7m 48s\n- openstack-ansible-deploy-aio_metal-debian-stable https://zuul.opendev.org/t/openstack/build/98ec43f003d248f3bab84fcf546fbf81 : SUCCESS in 1h 20m 44s\n- openstack-ansible-deploy-aio_metal-centos-7 https://zuul.opendev.org/t/openstack/build/10f6a092fa9c4d5bae167f17553cd064 : SUCCESS in 1h 40m 41s\n- openstack-ansible-deploy-aio_metal-opensuse-15 https://zuul.opendev.org/t/openstack/build/a2092668670e4f909ea776a802a1bc5e : FAILURE in 1h 27m 03s\n- openstack-ansible-deploy-aio_metal-ubuntu-bionic https://zuul.opendev.org/t/openstack/build/0fb45032acae4ffaa454afa4f60c15d6 : FAILURE in 1h 20m 21s\n- openstack-ansible-deploy-aio_distro_metal-centos-7 https://zuul.opendev.org/t/openstack/build/af474ba61883413e81f0130ee6921d9e : SUCCESS in 1h 18m 49s\n- openstack-ansible-deploy-aio_distro_metal-opensuse-15 https://zuul.opendev.org/t/openstack/build/14d4de9de94045d3820c84af907b3b11 : FAILURE in 28m 33s (non-voting)\n- openstack-ansible-deploy-aio_distro_metal-ubuntu-bionic https://zuul.opendev.org/t/openstack/build/be70f3475eb54b14aa18b59623255a30 : SUCCESS in 54m 07s\n- openstack-ansible-deploy-aio_telemetry_metal-centos-7 https://zuul.opendev.org/t/openstack/build/9cfabc6c14a346da9b00e210b48df30c : SUCCESS in 1h 42m 29s\n- openstack-ansible-deploy-aio_telemetry_metal-debian-stable https://zuul.opendev.org/t/openstack/build/3864736261d64cbbb4c7cdaeb3117d2e : SUCCESS in 1h 18m 53s\n- openstack-ansible-deploy-aio_telemetry_metal-ubuntu-bionic https://zuul.opendev.org/t/openstack/build/93d2998d8c4e4d34aa84a9739372b4f8 : SUCCESS in 1h 15m 39s\n- openstack-tox-docs https://zuul.opendev.org/t/openstack/build/c2af1ea119e2416ea871845db085786f : SUCCESS in 3m 15s","accounts_in_message":[],"_revision_number":1},{"id":"60e84fe8f6ba1eaad468927e893407ec30443d62","author":{"_account_id":28619,"name":"Dmitriy Rabotyagov","email":"noonedeadpunk@gmail.com","username":"noonedeadpunk"},"date":"2019-09-06 16:29:16.000000000","message":"Patch Set 1:\n\nrecheck","accounts_in_message":[],"_revision_number":1},{"id":"907f41628d4d79857508e85b39a5f71d49f187ef","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2019-09-06 18:47:07.000000000","message":"Patch Set 1: Verified+1\n\nBuild succeeded (check pipeline).\n\n- openstack-ansible-linters https://zuul.opendev.org/t/openstack/build/bb21cf5255ab46a7af423d6ecf89dea7 : SUCCESS in 13m 56s\n- openstack-ansible-deploy-aio_metal-debian-stable https://zuul.opendev.org/t/openstack/build/dae217222c2d4c30b0fa6156fc8f17fd : SUCCESS in 1h 21m 06s\n- openstack-ansible-deploy-aio_metal-centos-7 https://zuul.opendev.org/t/openstack/build/3b6c976c71764c90940d1cf61f19d4a9 : SUCCESS in 1h 53m 03s\n- openstack-ansible-deploy-aio_metal-opensuse-15 https://zuul.opendev.org/t/openstack/build/322655aeab4043e6ac2ca09a1ee54671 : SUCCESS in 1h 50m 20s\n- openstack-ansible-deploy-aio_metal-ubuntu-bionic https://zuul.opendev.org/t/openstack/build/bd068386a39048e0bc4e2a5dfb2eae33 : SUCCESS in 1h 26m 08s\n- openstack-ansible-deploy-aio_distro_metal-centos-7 https://zuul.opendev.org/t/openstack/build/0ef513fafd9c44139e4966b26c80db15 : SUCCESS in 1h 27m 15s\n- openstack-ansible-deploy-aio_distro_metal-opensuse-15 https://zuul.opendev.org/t/openstack/build/422504dab5b74b15bda822879f364742 : FAILURE in 32m 20s (non-voting)\n- openstack-ansible-deploy-aio_distro_metal-ubuntu-bionic https://zuul.opendev.org/t/openstack/build/a3a34241364f4edca8f57684a1536b6c : SUCCESS in 56m 28s\n- openstack-ansible-deploy-aio_telemetry_metal-centos-7 https://zuul.opendev.org/t/openstack/build/6b02f4707771423292c24ccfdd836900 : SUCCESS in 1h 57m 38s\n- openstack-ansible-deploy-aio_telemetry_metal-debian-stable https://zuul.opendev.org/t/openstack/build/48e2b727ef1b490fb984970e5596ad01 : SUCCESS in 1h 29m 56s\n- openstack-ansible-deploy-aio_telemetry_metal-ubuntu-bionic https://zuul.opendev.org/t/openstack/build/f96e56ddba3e433799c7346c57b0fae4 : SUCCESS in 1h 31m 17s\n- openstack-tox-docs https://zuul.opendev.org/t/openstack/build/0f6b3e46839541d1afdf33afa361de5d : SUCCESS in 3m 14s","accounts_in_message":[],"_revision_number":1},{"id":"91db76857d787287de812b16380780224823564b","author":{"_account_id":28619,"name":"Dmitriy Rabotyagov","email":"noonedeadpunk@gmail.com","username":"noonedeadpunk"},"date":"2019-09-06 20:38:47.000000000","message":"Patch Set 1: Code-Review+2 Workflow+1","accounts_in_message":[],"_revision_number":1},{"id":"87667e815603b7cf87da301b9fe96f6488ba4f34","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2019-09-06 20:38:59.000000000","message":"Patch Set 1: -Verified\n\nStarting gate jobs.","accounts_in_message":[],"_revision_number":1},{"id":"ff89fd37ce407b10cc03dfaaf8e82ce10b8970ab","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2019-09-06 22:52:25.000000000","message":"Change has been successfully merged by Zuul","accounts_in_message":[],"_revision_number":1},{"id":"4b6bf177441d3c8e959aab5c85b39b2265468945","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2019-09-06 22:52:25.000000000","message":"Patch Set 1: Verified+2\n\nBuild succeeded (gate pipeline).\n\n- openstack-ansible-linters https://zuul.opendev.org/t/openstack/build/8c2cedd1d6994edd85405be3d3819699 : SUCCESS in 7m 19s\n- openstack-ansible-deploy-aio_metal-debian-stable https://zuul.opendev.org/t/openstack/build/a8b630f63af6472f9f913b8dfbc7ef13 : SUCCESS in 1h 34m 43s\n- openstack-ansible-deploy-aio_metal-centos-7 https://zuul.opendev.org/t/openstack/build/f3fa171ac49a4de9bbabdf96cab8e8f5 : SUCCESS in 2h 07m 51s\n- openstack-ansible-deploy-aio_metal-opensuse-15 https://zuul.opendev.org/t/openstack/build/cc8c831fce284bdfb4e5f441cf411114 : SUCCESS in 1h 23m 38s\n- openstack-ansible-deploy-aio_metal-ubuntu-bionic https://zuul.opendev.org/t/openstack/build/c658b498ee414b37a277dd89d8efe45f : SUCCESS in 1h 27m 50s\n- openstack-ansible-deploy-aio_distro_metal-centos-7 https://zuul.opendev.org/t/openstack/build/fe25d46fd215456e8312b3c8fad61191 : SUCCESS in 1h 37m 18s\n- openstack-ansible-deploy-aio_distro_metal-ubuntu-bionic https://zuul.opendev.org/t/openstack/build/ae45c0fc78724c1fbdb3c494304786d3 : SUCCESS in 1h 07m 37s\n- openstack-ansible-deploy-aio_telemetry_metal-centos-7 https://zuul.opendev.org/t/openstack/build/59bcd783b3d242c8b2ea6efbbd7cc772 : SUCCESS in 1h 47m 43s\n- openstack-ansible-deploy-aio_telemetry_metal-debian-stable https://zuul.opendev.org/t/openstack/build/0fcd6270c8494165bbbb2f081f4b1e51 : SUCCESS in 1h 18m 02s\n- openstack-ansible-deploy-aio_telemetry_metal-ubuntu-bionic https://zuul.opendev.org/t/openstack/build/a659a5efb7854e6392dca4bc243be5b2 : SUCCESS in 1h 14m 32s\n- openstack-tox-docs https://zuul.opendev.org/t/openstack/build/a3093849608e4b2db96768b11d638bfe : SUCCESS in 3m 33s","accounts_in_message":[],"_revision_number":1},{"id":"6c6b85f40ca5cffc9502eee81e1739a0205c6c9b","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2019-09-06 22:54:06.000000000","message":"Patch Set 1:\n\nBuild succeeded (promote pipeline).\n\n- promote-openstack-tox-docs https://zuul.opendev.org/t/openstack/build/1fef2727c2fa4b48b42bace8e49bd564 : SUCCESS in 49s","accounts_in_message":[],"_revision_number":1}],"current_revision_number":1,"current_revision":"6ff08a4731e03874d07f523c2634b2236e9c9637","revisions":{"6ff08a4731e03874d07f523c2634b2236e9c9637":{"kind":"REWORK","_number":1,"created":"2019-09-04 15:11:48.000000000","uploader":{"_account_id":11131,"name":"OpenStack Proposal Bot","username":"proposal-bot","tags":["SERVICE_USER"]},"ref":"refs/changes/35/680035/1","fetch":{"anonymous http":{"url":"https://review.opendev.org/openstack/openstack-ansible-os_aodh","ref":"refs/changes/35/680035/1","commands":{"Checkout":"git fetch https://review.opendev.org/openstack/openstack-ansible-os_aodh refs/changes/35/680035/1 \u0026\u0026 git checkout FETCH_HEAD","Cherry Pick":"git fetch https://review.opendev.org/openstack/openstack-ansible-os_aodh refs/changes/35/680035/1 \u0026\u0026 git cherry-pick FETCH_HEAD","Format Patch":"git fetch https://review.opendev.org/openstack/openstack-ansible-os_aodh refs/changes/35/680035/1 \u0026\u0026 git format-patch -1 --stdout FETCH_HEAD","Pull":"git pull https://review.opendev.org/openstack/openstack-ansible-os_aodh refs/changes/35/680035/1"}}},"commit":{"parents":[{"commit":"d4422a137d2323eeddb4e278e288be3428f91011","subject":"Updated from OpenStack Ansible Tests","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/openstack-ansible-os_aodh/commit/d4422a137d2323eeddb4e278e288be3428f91011"}]}],"author":{"name":"OpenStack Proposal Bot","email":"openstack-infra@lists.openstack.org","date":"2019-09-04 15:11:43.000000000","tz":0},"committer":{"name":"OpenStack Proposal Bot","email":"openstack-infra@lists.openstack.org","date":"2019-09-04 15:11:43.000000000","tz":0},"subject":"Updated from OpenStack Ansible Tests","message":"Updated from OpenStack Ansible Tests\n\nChange-Id: Ie6ed1378f23645e71a80019f33b9ce482bb43680\n","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/openstack-ansible-os_aodh/commit/6ff08a4731e03874d07f523c2634b2236e9c9637"}],"resolve_conflicts_web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/openstack-ansible-os_aodh/commit/6ff08a4731e03874d07f523c2634b2236e9c9637"}]},"branch":"refs/heads/master"}},"requirements":[],"submit_records":[],"submit_requirements":[]}
