)]}'
{"id":"openstack%2Fopenstack-ansible-os_aodh~681408","triplet_id":"openstack%2Fopenstack-ansible-os_aodh~master~Ia5e78895c6bde67d680019055cfa8f96d38a26ee","project":"openstack/openstack-ansible-os_aodh","branch":"master","topic":"osa-py3","hashtags":[],"change_id":"Ia5e78895c6bde67d680019055cfa8f96d38a26ee","subject":"Allow venv python interpreter to be overridden","status":"MERGED","created":"2019-09-11 08:51:48.000000000","updated":"2019-09-12 05:20:29.000000000","submitted":"2019-09-12 05:18:59.000000000","submitter":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"total_comment_count":0,"unresolved_comment_count":0,"has_review_started":true,"submission_id":"681408-1568265539584-69ceffcb","meta_rev_id":"1e91ec4c1ec0af1c62ebef9eac20ba20e8b93e5f","_number":681408,"virtual_id_number":681408,"owner":{"_account_id":25023,"name":"Jonathan Rosser","email":"jonathan.rosser@rd.bbc.co.uk","username":"jrosser"},"actions":{},"labels":{"Verified":{"approved":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"all":[{"value":0,"_account_id":28008,"name":"Guilherme","display_name":"Guilherme Steinmüller","email":"gsteinmuller@vexxhost.com","username":"guilhermesp"},{"value":0,"_account_id":28619,"name":"Dmitriy Rabotyagov","email":"noonedeadpunk@gmail.com","username":"noonedeadpunk"},{"value":2,"date":"2019-09-12 05:18:59.000000000","permitted_voting_range":{"min":2,"max":2},"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},{"value":0,"date":"2019-09-11 12:38:01.000000000","_account_id":25023,"name":"Jonathan Rosser","email":"jonathan.rosser@rd.bbc.co.uk","username":"jrosser"}],"values":{"-2":"Fails","-1":"Doesn\u0027t seem to work"," 0":"No score","+1":"Works for me","+2":"Verified"},"description":"","default_value":0,"optional":true},"Code-Review":{"approved":{"_account_id":28008,"name":"Guilherme","display_name":"Guilherme Steinmüller","email":"gsteinmuller@vexxhost.com","username":"guilhermesp"},"all":[{"value":2,"date":"2019-09-12 02:49:47.000000000","permitted_voting_range":{"min":2,"max":2},"_account_id":28008,"name":"Guilherme","display_name":"Guilherme Steinmüller","email":"gsteinmuller@vexxhost.com","username":"guilhermesp"},{"value":2,"date":"2019-09-11 16:46:07.000000000","permitted_voting_range":{"min":2,"max":2},"_account_id":28619,"name":"Dmitriy Rabotyagov","email":"noonedeadpunk@gmail.com","username":"noonedeadpunk"},{"value":0,"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},{"value":0,"_account_id":25023,"name":"Jonathan Rosser","email":"jonathan.rosser@rd.bbc.co.uk","username":"jrosser"}],"values":{"-2":"Do not merge","-1":"This patch needs further work before it can be merged"," 0":"No score","+1":"Looks good to me, but someone else must approve","+2":"Looks good to me (core reviewer)"},"description":"","default_value":0,"optional":true},"Workflow":{"approved":{"_account_id":28008,"name":"Guilherme","display_name":"Guilherme Steinmüller","email":"gsteinmuller@vexxhost.com","username":"guilhermesp"},"all":[{"value":1,"date":"2019-09-12 02:49:47.000000000","permitted_voting_range":{"min":1,"max":1},"_account_id":28008,"name":"Guilherme","display_name":"Guilherme Steinmüller","email":"gsteinmuller@vexxhost.com","username":"guilhermesp"},{"value":0,"_account_id":28619,"name":"Dmitriy Rabotyagov","email":"noonedeadpunk@gmail.com","username":"noonedeadpunk"},{"value":0,"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},{"value":0,"_account_id":25023,"name":"Jonathan Rosser","email":"jonathan.rosser@rd.bbc.co.uk","username":"jrosser"}],"values":{"-1":"Work in progress"," 0":"Ready for reviews","+1":"Approved"},"description":"","default_value":0,"optional":true},"Backport-Candidate":{"all":[{"value":0,"_account_id":28008,"name":"Guilherme","display_name":"Guilherme Steinmüller","email":"gsteinmuller@vexxhost.com","username":"guilhermesp"},{"value":0,"_account_id":28619,"name":"Dmitriy Rabotyagov","email":"noonedeadpunk@gmail.com","username":"noonedeadpunk"},{"value":0,"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},{"value":0,"_account_id":25023,"name":"Jonathan Rosser","email":"jonathan.rosser@rd.bbc.co.uk","username":"jrosser"}],"values":{"-1":"Not A Backport Candidate"," 0":"Backport Review Needed","+1":"Proposed Backport"},"description":"","default_value":0,"optional":true}},"removable_reviewers":[],"reviewers":{"REVIEWER":[{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},{"_account_id":25023,"name":"Jonathan Rosser","email":"jonathan.rosser@rd.bbc.co.uk","username":"jrosser"},{"_account_id":28008,"name":"Guilherme","display_name":"Guilherme Steinmüller","email":"gsteinmuller@vexxhost.com","username":"guilhermesp"},{"_account_id":28619,"name":"Dmitriy Rabotyagov","email":"noonedeadpunk@gmail.com","username":"noonedeadpunk"}]},"pending_reviewers":{},"reviewer_updates":[{"updated":"2019-09-11 16:46:07.000000000","updated_by":{"_account_id":28619,"name":"Dmitriy Rabotyagov","email":"noonedeadpunk@gmail.com","username":"noonedeadpunk"},"reviewer":{"_account_id":28619,"name":"Dmitriy Rabotyagov","email":"noonedeadpunk@gmail.com","username":"noonedeadpunk"},"state":"REVIEWER"},{"updated":"2019-09-12 02:49:47.000000000","updated_by":{"_account_id":28008,"name":"Guilherme","display_name":"Guilherme Steinmüller","email":"gsteinmuller@vexxhost.com","username":"guilhermesp"},"reviewer":{"_account_id":28008,"name":"Guilherme","display_name":"Guilherme Steinmüller","email":"gsteinmuller@vexxhost.com","username":"guilhermesp"},"state":"REVIEWER"},{"updated":"2019-09-12 05:18:59.000000000","updated_by":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"reviewer":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"state":"REVIEWER"}],"messages":[{"id":"9b4042ebeceb679b8377341949533f37aa7b60f1","author":{"_account_id":25023,"name":"Jonathan Rosser","email":"jonathan.rosser@rd.bbc.co.uk","username":"jrosser"},"date":"2019-09-11 08:51:48.000000000","message":"Uploaded patch set 1.","accounts_in_message":[],"_revision_number":1},{"id":"e0ecfb7741a696a30ce38215c350cf060277345f","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2019-09-11 11:07:35.000000000","message":"Patch Set 1: Verified-1\n\nBuild failed (check pipeline).  For information on how to proceed, see\nhttp://docs.openstack.org/infra/manual/developers.html#automated-testing\n\n\n- openstack-ansible-linters https://zuul.opendev.org/t/openstack/build/60d4c3d513e94c9091729fbb8d56c211 : SUCCESS in 5m 40s\n- openstack-ansible-deploy-aio_metal-debian-stable https://zuul.opendev.org/t/openstack/build/4b42e7cb6b5e4f87adea2dcd98db764a : SUCCESS in 1h 42m 56s\n- openstack-ansible-deploy-aio_metal-centos-7 https://zuul.opendev.org/t/openstack/build/59edb7f51512441992d49cda7ffe871b : SUCCESS in 2h 02m 47s\n- openstack-ansible-deploy-aio_metal-opensuse-15 https://zuul.opendev.org/t/openstack/build/1a4f9553b879434cb5ada7474e48f1d0 : FAILURE in 30m 08s\n- openstack-ansible-deploy-aio_metal-ubuntu-bionic https://zuul.opendev.org/t/openstack/build/97ef78ba5af94452b4dcb0f8ca3691d2 : SUCCESS in 1h 29m 17s\n- openstack-ansible-deploy-aio_distro_metal-centos-7 https://zuul.opendev.org/t/openstack/build/d5fc51276f4a4cd795100b89b5e95336 : SUCCESS in 1h 27m 05s\n- openstack-ansible-deploy-aio_distro_metal-opensuse-15 https://zuul.opendev.org/t/openstack/build/80a4d886fabf4c16863e2ec9f5dcbe8d : FAILURE in 34m 09s (non-voting)\n- openstack-ansible-deploy-aio_distro_metal-ubuntu-bionic https://zuul.opendev.org/t/openstack/build/b5747a2d54ef492cbd570fd2c325edb0 : SUCCESS in 1h 12m 08s\n- openstack-ansible-deploy-aio_telemetry_metal-centos-7 https://zuul.opendev.org/t/openstack/build/9346f6d1b090484892c7be259d3a1ab0 : SUCCESS in 1h 56m 29s\n- openstack-ansible-deploy-aio_telemetry_metal-debian-stable https://zuul.opendev.org/t/openstack/build/d39f19aa0c60488f873eca739222ba14 : SUCCESS in 1h 41m 27s\n- openstack-ansible-deploy-aio_telemetry_metal-ubuntu-bionic https://zuul.opendev.org/t/openstack/build/14e641981dd042cd96e8c7a30c035581 : SUCCESS in 2h 11m 15s\n- openstack-tox-docs https://zuul.opendev.org/t/openstack/build/6744e39dc02942e6a322e3e271e74e4e : SUCCESS in 3m 02s","accounts_in_message":[],"_revision_number":1},{"id":"0e6150b5146ff0417a80a5082c43109d14a2f82f","author":{"_account_id":25023,"name":"Jonathan Rosser","email":"jonathan.rosser@rd.bbc.co.uk","username":"jrosser"},"date":"2019-09-11 12:38:01.000000000","message":"Patch Set 1:\n\nrecheck","accounts_in_message":[],"_revision_number":1},{"id":"60129afdb464f5a3c1214b870b75d9b14c849bfc","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2019-09-11 15:01:26.000000000","message":"Patch Set 1: Verified+1\n\nBuild succeeded (check pipeline).\n\n- openstack-ansible-linters https://zuul.opendev.org/t/openstack/build/7b45e3df89b6461794ee57dd9fe04dc9 : SUCCESS in 4m 49s\n- openstack-ansible-deploy-aio_metal-debian-stable https://zuul.opendev.org/t/openstack/build/9b0772a33a5a4d20b731ccbb0c05146d : SUCCESS in 2h 03m 43s\n- openstack-ansible-deploy-aio_metal-centos-7 https://zuul.opendev.org/t/openstack/build/32311f8f4f37440aa895d925f7fe1300 : SUCCESS in 2h 17m 04s\n- openstack-ansible-deploy-aio_metal-opensuse-15 https://zuul.opendev.org/t/openstack/build/7d8434923e1248b8a02e3bcaaace2cd7 : SUCCESS in 1h 49m 38s\n- openstack-ansible-deploy-aio_metal-ubuntu-bionic https://zuul.opendev.org/t/openstack/build/2bbc50301c814fc3afab53b7d50480aa : SUCCESS in 1h 29m 56s\n- openstack-ansible-deploy-aio_distro_metal-centos-7 https://zuul.opendev.org/t/openstack/build/7728b32f05c44e59a9fe1f0be10e155e : SUCCESS in 1h 22m 41s\n- openstack-ansible-deploy-aio_distro_metal-opensuse-15 https://zuul.opendev.org/t/openstack/build/aafe2fd2f8db45a583dffdc9c814ae60 : FAILURE in 33m 11s (non-voting)\n- openstack-ansible-deploy-aio_distro_metal-ubuntu-bionic https://zuul.opendev.org/t/openstack/build/f622e5cac344483296704306d985ee90 : SUCCESS in 1h 05m 31s\n- openstack-ansible-deploy-aio_telemetry_metal-centos-7 https://zuul.opendev.org/t/openstack/build/3cd7440fc108457f807b081088ede57e : SUCCESS in 1h 55m 17s\n- openstack-ansible-deploy-aio_telemetry_metal-debian-stable https://zuul.opendev.org/t/openstack/build/4b2e0bb8ec6a44fbb61f08a82702387c : SUCCESS in 1h 31m 51s\n- openstack-ansible-deploy-aio_telemetry_metal-ubuntu-bionic https://zuul.opendev.org/t/openstack/build/ce138fa4e4ea4fc09ba843c89a6b06ad : SUCCESS in 1h 27m 55s\n- openstack-tox-docs https://zuul.opendev.org/t/openstack/build/0c67c7c9fcc147e5bca551aedcd472c7 : SUCCESS in 4m 25s","accounts_in_message":[],"_revision_number":1},{"id":"651697f92874422911d9318af567b4b62fb8b10e","author":{"_account_id":28619,"name":"Dmitriy Rabotyagov","email":"noonedeadpunk@gmail.com","username":"noonedeadpunk"},"date":"2019-09-11 16:46:07.000000000","message":"Patch Set 1: Code-Review+2","accounts_in_message":[],"_revision_number":1},{"id":"bbbb141ae641ee584af1939afa11b7e90670cd7e","author":{"_account_id":28008,"name":"Guilherme","display_name":"Guilherme Steinmüller","email":"gsteinmuller@vexxhost.com","username":"guilhermesp"},"date":"2019-09-12 02:49:47.000000000","message":"Patch Set 1: Code-Review+2 Workflow+1","accounts_in_message":[],"_revision_number":1},{"id":"34540d2c70c45d6e5e48e76ebfd94aa475b4bf38","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2019-09-12 02:49:58.000000000","message":"Patch Set 1: -Verified\n\nStarting gate jobs.","accounts_in_message":[],"_revision_number":1},{"id":"94956e708d9a1a88937dbb4d85a710b894a8e42f","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2019-09-12 05:18:59.000000000","message":"Change has been successfully merged by Zuul","accounts_in_message":[],"_revision_number":1},{"id":"a2f962d89bc5c88ac6cef3e041163b0d290feda9","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2019-09-12 05:18:59.000000000","message":"Patch Set 1: Verified+2\n\nBuild succeeded (gate pipeline).\n\n- openstack-ansible-linters https://zuul.opendev.org/t/openstack/build/abe07c7fbf7b4f3e8e9c4211bc1bdfbe : SUCCESS in 5m 42s\n- openstack-ansible-deploy-aio_metal-debian-stable https://zuul.opendev.org/t/openstack/build/faba20b0f72c4dcdbfa93213904a6112 : SUCCESS in 1h 33m 56s\n- openstack-ansible-deploy-aio_metal-centos-7 https://zuul.opendev.org/t/openstack/build/b30c701418b94bf0a8eb66e497ef4c30 : SUCCESS in 2h 22m 04s\n- openstack-ansible-deploy-aio_metal-opensuse-15 https://zuul.opendev.org/t/openstack/build/0f470114f6c0415cb977f49389eb488a : SUCCESS in 1h 44m 18s\n- openstack-ansible-deploy-aio_metal-ubuntu-bionic https://zuul.opendev.org/t/openstack/build/c7bbc7bb842c48128dbf460e43fa5bdd : SUCCESS in 1h 24m 00s\n- openstack-ansible-deploy-aio_distro_metal-centos-7 https://zuul.opendev.org/t/openstack/build/f9d5071538794c2a8fa06ca86eb381bf : SUCCESS in 1h 54m 02s\n- openstack-ansible-deploy-aio_distro_metal-ubuntu-bionic https://zuul.opendev.org/t/openstack/build/9fd93a42325446d59e8612c9ee604a78 : SUCCESS in 1h 10m 46s\n- openstack-ansible-deploy-aio_telemetry_metal-centos-7 https://zuul.opendev.org/t/openstack/build/75a2e6c2198244598dfe0c333592748f : SUCCESS in 2h 20m 31s\n- openstack-ansible-deploy-aio_telemetry_metal-debian-stable https://zuul.opendev.org/t/openstack/build/be09530ffbac4207a5f6ec12183b1813 : SUCCESS in 1h 33m 40s\n- openstack-ansible-deploy-aio_telemetry_metal-ubuntu-bionic https://zuul.opendev.org/t/openstack/build/0565308aa3dc42bc85b5228dd29e588e : SUCCESS in 1h 42m 58s\n- openstack-tox-docs https://zuul.opendev.org/t/openstack/build/a42924b55ea04f9a942595559e7ba9bf : SUCCESS in 4m 19s","accounts_in_message":[],"_revision_number":1},{"id":"38e95adf9c445bc888d57190394b69a0569311e0","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2019-09-12 05:20:29.000000000","message":"Patch Set 1:\n\nBuild succeeded (promote pipeline).\n\n- promote-openstack-tox-docs https://zuul.opendev.org/t/openstack/build/1157b14f5c5f4f96bc7738858fe9a882 : SUCCESS in 1m 10s","accounts_in_message":[],"_revision_number":1}],"current_revision_number":1,"current_revision":"6fca9b86cb3f697f482a5b334d2faed5742a406b","revisions":{"6fca9b86cb3f697f482a5b334d2faed5742a406b":{"kind":"REWORK","_number":1,"created":"2019-09-11 08:51:48.000000000","uploader":{"_account_id":25023,"name":"Jonathan Rosser","email":"jonathan.rosser@rd.bbc.co.uk","username":"jrosser"},"ref":"refs/changes/08/681408/1","fetch":{"anonymous http":{"url":"https://review.opendev.org/openstack/openstack-ansible-os_aodh","ref":"refs/changes/08/681408/1","commands":{"Checkout":"git fetch https://review.opendev.org/openstack/openstack-ansible-os_aodh refs/changes/08/681408/1 \u0026\u0026 git checkout FETCH_HEAD","Cherry Pick":"git fetch https://review.opendev.org/openstack/openstack-ansible-os_aodh refs/changes/08/681408/1 \u0026\u0026 git cherry-pick FETCH_HEAD","Format Patch":"git fetch https://review.opendev.org/openstack/openstack-ansible-os_aodh refs/changes/08/681408/1 \u0026\u0026 git format-patch -1 --stdout FETCH_HEAD","Pull":"git pull https://review.opendev.org/openstack/openstack-ansible-os_aodh refs/changes/08/681408/1"}}},"commit":{"parents":[{"commit":"ef51898fef8cd7529bfbf9c43ba7bb66f55d166a","subject":"Merge \"Migrate aodh to use uWSGI role\"","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/openstack-ansible-os_aodh/commit/ef51898fef8cd7529bfbf9c43ba7bb66f55d166a"}]}],"author":{"name":"Jonathan Rosser","email":"jonathan.rosser@rd.bbc.co.uk","date":"2019-09-11 08:51:13.000000000","tz":60},"committer":{"name":"Jonathan Rosser","email":"jonathan.rosser@rd.bbc.co.uk","date":"2019-09-11 08:51:13.000000000","tz":60},"subject":"Allow venv python interpreter to be overridden","message":"Allow venv python interpreter to be overridden\n\nChange-Id: Ia5e78895c6bde67d680019055cfa8f96d38a26ee\n","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/openstack-ansible-os_aodh/commit/6fca9b86cb3f697f482a5b334d2faed5742a406b"}],"resolve_conflicts_web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/openstack-ansible-os_aodh/commit/6fca9b86cb3f697f482a5b334d2faed5742a406b"}]},"branch":"refs/heads/master"}},"requirements":[],"submit_records":[],"submit_requirements":[]}
