)]}'
{"id":"openstack%2Fopenstack-ansible-os_aodh~690346","triplet_id":"openstack%2Fopenstack-ansible-os_aodh~stable%2Ftrain~I53997de1aab611a3627a861fdebacaf66c39688b","project":"openstack/openstack-ansible-os_aodh","branch":"stable/train","topic":"openstack/openstack-ansible-tests/sync-tests","hashtags":[],"change_id":"I53997de1aab611a3627a861fdebacaf66c39688b","subject":"Updated from OpenStack Ansible Tests","status":"MERGED","created":"2019-10-22 19:15:55.000000000","updated":"2020-04-06 19:14:20.000000000","submitted":"2020-04-06 19:12:40.000000000","submitter":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"total_comment_count":0,"unresolved_comment_count":0,"has_review_started":true,"submission_id":"690346-1586200361059-28344ecc","meta_rev_id":"84e8821f776d8da6bf084c42249b3c0ad37493f8","_number":690346,"virtual_id_number":690346,"owner":{"_account_id":11131,"name":"OpenStack Proposal Bot","username":"proposal-bot","tags":["SERVICE_USER"]},"actions":{},"labels":{"Verified":{"approved":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"all":[{"value":0,"_account_id":28619,"name":"Dmitriy Rabotyagov","email":"noonedeadpunk@gmail.com","username":"noonedeadpunk"},{"tag":"autogenerated:zuul:gate","value":2,"date":"2020-04-06 19:12:40.000000000","permitted_voting_range":{"min":2,"max":2},"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},{"value":0,"date":"2020-04-02 14:03:04.000000000","_account_id":25023,"name":"Jonathan Rosser","email":"jonathan.rosser@rd.bbc.co.uk","username":"jrosser"}],"values":{"-2":"Fails","-1":"Doesn\u0027t seem to work"," 0":"No score","+1":"Works for me","+2":"Verified"},"description":"","default_value":0,"optional":true},"Code-Review":{"approved":{"_account_id":28619,"name":"Dmitriy Rabotyagov","email":"noonedeadpunk@gmail.com","username":"noonedeadpunk"},"all":[{"value":2,"date":"2020-03-24 19:47:18.000000000","permitted_voting_range":{"min":2,"max":2},"_account_id":28619,"name":"Dmitriy Rabotyagov","email":"noonedeadpunk@gmail.com","username":"noonedeadpunk"},{"value":0,"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},{"value":0,"_account_id":25023,"name":"Jonathan Rosser","email":"jonathan.rosser@rd.bbc.co.uk","username":"jrosser"}],"values":{"-2":"Do not merge","-1":"This patch needs further work before it can be merged"," 0":"No score","+1":"Looks good to me, but someone else must approve","+2":"Looks good to me (core reviewer)"},"description":"","default_value":0,"optional":true},"Workflow":{"approved":{"_account_id":28619,"name":"Dmitriy Rabotyagov","email":"noonedeadpunk@gmail.com","username":"noonedeadpunk"},"all":[{"value":1,"date":"2020-03-24 19:47:18.000000000","permitted_voting_range":{"min":1,"max":1},"_account_id":28619,"name":"Dmitriy Rabotyagov","email":"noonedeadpunk@gmail.com","username":"noonedeadpunk"},{"value":0,"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},{"value":0,"_account_id":25023,"name":"Jonathan Rosser","email":"jonathan.rosser@rd.bbc.co.uk","username":"jrosser"}],"values":{"-1":"Work in progress"," 0":"Ready for reviews","+1":"Approved"},"description":"","default_value":0,"optional":true},"Backport-Candidate":{"all":[{"value":0,"_account_id":28619,"name":"Dmitriy Rabotyagov","email":"noonedeadpunk@gmail.com","username":"noonedeadpunk"},{"value":0,"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},{"value":0,"_account_id":25023,"name":"Jonathan Rosser","email":"jonathan.rosser@rd.bbc.co.uk","username":"jrosser"}],"values":{"-1":"Not A Backport Candidate"," 0":"Backport Review Needed","+1":"Proposed Backport"},"description":"","default_value":0,"optional":true}},"removable_reviewers":[],"reviewers":{"REVIEWER":[{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},{"_account_id":25023,"name":"Jonathan Rosser","email":"jonathan.rosser@rd.bbc.co.uk","username":"jrosser"},{"_account_id":28619,"name":"Dmitriy Rabotyagov","email":"noonedeadpunk@gmail.com","username":"noonedeadpunk"}]},"pending_reviewers":{},"reviewer_updates":[{"updated":"2020-03-24 19:47:18.000000000","updated_by":{"_account_id":28619,"name":"Dmitriy Rabotyagov","email":"noonedeadpunk@gmail.com","username":"noonedeadpunk"},"reviewer":{"_account_id":28619,"name":"Dmitriy Rabotyagov","email":"noonedeadpunk@gmail.com","username":"noonedeadpunk"},"state":"REVIEWER"},{"updated":"2020-04-02 14:03:04.000000000","updated_by":{"_account_id":25023,"name":"Jonathan Rosser","email":"jonathan.rosser@rd.bbc.co.uk","username":"jrosser"},"reviewer":{"_account_id":25023,"name":"Jonathan Rosser","email":"jonathan.rosser@rd.bbc.co.uk","username":"jrosser"},"state":"REVIEWER"},{"updated":"2020-04-06 19:12:40.000000000","updated_by":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"reviewer":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"state":"REVIEWER"}],"messages":[{"id":"2d0be6e9333e2609f08548167da5947758e63271","author":{"_account_id":11131,"name":"OpenStack Proposal Bot","username":"proposal-bot","tags":["SERVICE_USER"]},"date":"2019-10-22 19:15:55.000000000","message":"Uploaded patch set 1.","accounts_in_message":[],"_revision_number":1},{"id":"dac1d1a824e57c3561b9f9bb02a5b622215e87ac","tag":"autogenerated:zuul:check","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2019-10-22 19:30:44.000000000","message":"Patch Set 1: Verified-1\n\nBuild failed (check pipeline).  For information on how to proceed, see\nhttp://docs.openstack.org/infra/manual/developers.html#automated-testing\n\n\n- requirements-check https://zuul.opendev.org/t/openstack/build/5c529b44d5d7473fab23c00925c3582a : SUCCESS in 3m 51s\n- openstack-ansible-linters https://zuul.opendev.org/t/openstack/build/fe6b7430e1d04180911faf2594626d79 : SUCCESS in 7m 35s\n- openstack-tox-docs https://zuul.opendev.org/t/openstack/build/6226662a4c5b4ac5aa96d1ddba894c72 : FAILURE in 5m 35s\n- build-openstack-releasenotes https://zuul.opendev.org/t/openstack/build/79ad333fdaaf4621af58dd5aa29be113 : SUCCESS in 2m 53s","accounts_in_message":[],"_revision_number":1},{"id":"abf0e73b2cacfa902b0ddc323df5bc57a2e644b0","author":{"_account_id":11131,"name":"OpenStack Proposal Bot","username":"proposal-bot","tags":["SERVICE_USER"]},"date":"2020-03-24 18:33:30.000000000","message":"Uploaded patch set 2.","accounts_in_message":[],"_revision_number":2},{"id":"861093ef8049f83b1258c1d4613f581502d75a5e","author":{"_account_id":28619,"name":"Dmitriy Rabotyagov","email":"noonedeadpunk@gmail.com","username":"noonedeadpunk"},"date":"2020-03-24 19:47:18.000000000","message":"Patch Set 2: Code-Review+2 Workflow+1","accounts_in_message":[],"_revision_number":2},{"id":"56197f635cb36b8d9c2fb32d0e4114cc075c9f4c","tag":"autogenerated:zuul:check","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2020-03-24 22:16:27.000000000","message":"Patch Set 2: Verified-1\n\nBuild failed (check pipeline).  For information on how to proceed, see\nhttps://docs.opendev.org/opendev/infra-manual/latest/developers.html#automated-testing\n\n\n- openstack-ansible-linters https://zuul.opendev.org/t/openstack/build/71b92a6b04dc45b98d8febae52a16e71 : SUCCESS in 8m 15s\n- openstack-ansible-deploy-aio_metal-debian-buster https://zuul.opendev.org/t/openstack/build/d0c443c01e094e7eaea194001e026513 : SUCCESS in 1h 20m 38s\n- openstack-ansible-deploy-aio_metal-debian-stable https://zuul.opendev.org/t/openstack/build/6fe4da24f971498585a1b38901656885 : SUCCESS in 1h 32m 07s\n- openstack-ansible-deploy-aio_metal-centos-7 https://zuul.opendev.org/t/openstack/build/fbbcff01aabc43c0a4b649be04b3e2e3 : RETRY_LIMIT in 13m 42s\n- openstack-ansible-deploy-aio_metal-opensuse-15 https://zuul.opendev.org/t/openstack/build/43054ac8deb047ccba98c96417ee8131 : SUCCESS in 1h 39m 05s\n- openstack-ansible-deploy-aio_metal-ubuntu-bionic https://zuul.opendev.org/t/openstack/build/b534293774fe41bebb6f0cffd1fa7f65 : SUCCESS in 1h 33m 44s\n- openstack-ansible-upgrade-aio_metal-centos-7 https://zuul.opendev.org/t/openstack/build/587f195379cb4f6ea031e21d8dcf4f32 : SUCCESS in 2h 52m 52s (non-voting)\n- openstack-ansible-upgrade-aio_metal-ubuntu-bionic https://zuul.opendev.org/t/openstack/build/a8cff41cdd4049e0880104c143d7aa21 : SUCCESS in 2h 10m 55s (non-voting)\n- openstack-ansible-deploy-aio_distro_metal-centos-7 https://zuul.opendev.org/t/openstack/build/d15be42c02b14204a055c448f0d2ea90 : RETRY_LIMIT in 9m 57s\n- openstack-ansible-deploy-aio_distro_metal-opensuse-15 https://zuul.opendev.org/t/openstack/build/898365cbe59b4166a26836400ec20d23 : FAILURE in 29m 12s (non-voting)\n- openstack-ansible-deploy-aio_distro_metal-ubuntu-bionic https://zuul.opendev.org/t/openstack/build/725cacb0484a4c29a32e0182434d5815 : SUCCESS in 1h 17m 36s\n- openstack-ansible-deploy-aio_telemetry_metal-centos-7 https://zuul.opendev.org/t/openstack/build/cffd3d1786464693b76d4681a844d312 : RETRY_LIMIT in 9m 22s\n- openstack-ansible-deploy-aio_telemetry_metal-debian-buster https://zuul.opendev.org/t/openstack/build/140b530d34b84305b80233763810c4ae : SUCCESS in 1h 23m 29s\n- openstack-ansible-deploy-aio_telemetry_metal-debian-stable https://zuul.opendev.org/t/openstack/build/922ff8455ae744f5822a1037d412b520 : SUCCESS in 1h 33m 05s\n- openstack-ansible-deploy-aio_telemetry_metal-ubuntu-bionic https://zuul.opendev.org/t/openstack/build/9ded84dbb197486ea023ebec9a7e45a1 : SUCCESS in 1h 42m 01s\n- openstack-tox-docs https://zuul.opendev.org/t/openstack/build/23e83989b708490e9841bfc1a8f1f966 : SUCCESS in 11m 07s\n- build-openstack-releasenotes https://zuul.opendev.org/t/openstack/build/3680e2409af040288f9d9fe80400ec23 : SUCCESS in 5m 19s","accounts_in_message":[],"_revision_number":2},{"id":"dc6593ca6bec6c51895a182c686cb951a6403da4","author":{"_account_id":25023,"name":"Jonathan Rosser","email":"jonathan.rosser@rd.bbc.co.uk","username":"jrosser"},"date":"2020-03-28 08:02:54.000000000","message":"Patch Set 2:\n\nrecheck","accounts_in_message":[],"_revision_number":2},{"id":"e0c15a0de1a670e30eacf593332cfedfa933f816","tag":"autogenerated:zuul:check","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2020-03-28 11:12:29.000000000","message":"Patch Set 2:\n\nBuild failed (check pipeline).  For information on how to proceed, see\nhttps://docs.opendev.org/opendev/infra-manual/latest/developers.html#automated-testing\n\n\n- openstack-ansible-linters https://zuul.opendev.org/t/openstack/build/0ee0c0c9324e4753b2a86090236db4cc : SUCCESS in 6m 31s\n- openstack-ansible-deploy-aio_metal-debian-buster https://zuul.opendev.org/t/openstack/build/97f89a5a17974df18d5e63acac31d024 : SUCCESS in 1h 18m 04s\n- openstack-ansible-deploy-aio_metal-debian-stable https://zuul.opendev.org/t/openstack/build/0d3e331896c8448ebe255a5e35d27d56 : FAILURE in 2h 17m 43s\n- openstack-ansible-deploy-aio_metal-centos-7 https://zuul.opendev.org/t/openstack/build/cd06331bfc974eb59e430326a1a9506e : RETRY_LIMIT in 6m 36s\n- openstack-ansible-deploy-aio_metal-opensuse-15 https://zuul.opendev.org/t/openstack/build/d1886a3ca63945cc92001c80b611eaa3 : SUCCESS in 2h 16m 24s\n- openstack-ansible-deploy-aio_metal-ubuntu-bionic https://zuul.opendev.org/t/openstack/build/92f2a3df59514175a1852a14f7cd5ee2 : SUCCESS in 1h 20m 27s\n- openstack-ansible-upgrade-aio_metal-centos-7 https://zuul.opendev.org/t/openstack/build/4fff3dd7b64b4c6f90cd4987842a7a05 : FAILURE in 3h 01m 54s (non-voting)\n- openstack-ansible-upgrade-aio_metal-ubuntu-bionic https://zuul.opendev.org/t/openstack/build/55e1e661250f4539b8c16d73ecf961d5 : SUCCESS in 2h 15m 20s (non-voting)\n- openstack-ansible-deploy-aio_distro_metal-centos-7 https://zuul.opendev.org/t/openstack/build/9d27a2dfd4844bc59c4342c23b2c25c3 : RETRY_LIMIT in 6m 57s\n- openstack-ansible-deploy-aio_distro_metal-opensuse-15 https://zuul.opendev.org/t/openstack/build/1591543a56484e24aec0bfeebac49af3 : FAILURE in 33m 23s (non-voting)\n- openstack-ansible-deploy-aio_distro_metal-ubuntu-bionic https://zuul.opendev.org/t/openstack/build/63c7e6bd9be74e1da9e0a3374e527c2c : SUCCESS in 1h 06m 03s\n- openstack-ansible-deploy-aio_telemetry_metal-centos-7 https://zuul.opendev.org/t/openstack/build/44af9e1fbf924fc590eacf762d82fd4e : RETRY_LIMIT in 9m 10s\n- openstack-ansible-deploy-aio_telemetry_metal-debian-buster https://zuul.opendev.org/t/openstack/build/3364471d6fc54544a1c6183befa315e9 : SUCCESS in 1h 16m 16s\n- openstack-ansible-deploy-aio_telemetry_metal-debian-stable https://zuul.opendev.org/t/openstack/build/a7903e4df3cb43a0ab2e7376210b081a : FAILURE in 1h 27m 13s\n- openstack-ansible-deploy-aio_telemetry_metal-ubuntu-bionic https://zuul.opendev.org/t/openstack/build/88a36d1222824863a2ae6e0040eb9bb9 : SUCCESS in 1h 27m 07s\n- openstack-tox-docs https://zuul.opendev.org/t/openstack/build/e2f11b91ca204f97b5c706371304dbbc : SUCCESS in 5m 30s\n- build-openstack-releasenotes https://zuul.opendev.org/t/openstack/build/0ab86b6820b74f19b238e012084813ad : SUCCESS in 3m 31s","accounts_in_message":[],"_revision_number":2},{"id":"535579395a83887677bd420d22981be8a4e70fc7","author":{"_account_id":25023,"name":"Jonathan Rosser","email":"jonathan.rosser@rd.bbc.co.uk","username":"jrosser"},"date":"2020-04-02 14:03:04.000000000","message":"Patch Set 2:\n\nrecheck","accounts_in_message":[],"_revision_number":2},{"id":"bea7490d8a69c4de5cad70206df2308850444dfc","tag":"autogenerated:zuul:check","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2020-04-02 17:09:34.000000000","message":"Patch Set 2:\n\nBuild failed (check pipeline).  For information on how to proceed, see\nhttps://docs.opendev.org/opendev/infra-manual/latest/developers.html#automated-testing\n\n\n- openstack-ansible-linters https://zuul.opendev.org/t/openstack/build/56b3d631ad494ee9a0990c5a06910b31 : SUCCESS in 8m 42s\n- openstack-ansible-deploy-aio_metal-debian-buster https://zuul.opendev.org/t/openstack/build/4f8c1ff3fdbf40508677590d99e2e408 : SUCCESS in 1h 21m 29s\n- openstack-ansible-deploy-aio_metal-debian-stable https://zuul.opendev.org/t/openstack/build/88ec54c2c0154ed2bb93a73799ce5222 : SUCCESS in 1h 31m 51s\n- openstack-ansible-deploy-aio_metal-centos-7 https://zuul.opendev.org/t/openstack/build/92fcf322200141058479f65a0ea59e20 : SUCCESS in 2h 00m 59s\n- openstack-ansible-deploy-aio_metal-opensuse-15 https://zuul.opendev.org/t/openstack/build/d1deb58027d74afd813b6b7b5ad49478 : FAILURE in 24m 37s\n- openstack-ansible-deploy-aio_metal-ubuntu-bionic https://zuul.opendev.org/t/openstack/build/60eb3d242e4c41ed9b6fb4b655083286 : SUCCESS in 1h 29m 50s\n- openstack-ansible-upgrade-aio_metal-centos-7 https://zuul.opendev.org/t/openstack/build/2f7286cd5e324c9ab95db745d38f98f2 : SUCCESS in 2h 53m 29s (non-voting)\n- openstack-ansible-upgrade-aio_metal-ubuntu-bionic https://zuul.opendev.org/t/openstack/build/9790748b73e649da8449bca17093370f : SUCCESS in 2h 24m 43s (non-voting)\n- openstack-ansible-deploy-aio_distro_metal-centos-7 https://zuul.opendev.org/t/openstack/build/8a11e49dc28b4b59b881b749e96816ce : SUCCESS in 1h 32m 36s\n- openstack-ansible-deploy-aio_distro_metal-opensuse-15 https://zuul.opendev.org/t/openstack/build/59f0518b496041b9a706a08a906aefe1 : FAILURE in 34m 31s (non-voting)\n- openstack-ansible-deploy-aio_distro_metal-ubuntu-bionic https://zuul.opendev.org/t/openstack/build/f8a0dddc0f8a45eab76ff6942c1b9b6a : SUCCESS in 1h 08m 23s\n- openstack-ansible-deploy-aio_telemetry_metal-centos-7 https://zuul.opendev.org/t/openstack/build/189ce16558f040f69588f6eca410f8d5 : SUCCESS in 2h 02m 46s\n- openstack-ansible-deploy-aio_telemetry_metal-debian-buster https://zuul.opendev.org/t/openstack/build/2611eab2586c4b409c243c2ac5b28294 : SUCCESS in 1h 21m 24s\n- openstack-ansible-deploy-aio_telemetry_metal-debian-stable https://zuul.opendev.org/t/openstack/build/eb2065d3adf243f4a9c2b8381cb6ab13 : SUCCESS in 1h 44m 49s\n- openstack-ansible-deploy-aio_telemetry_metal-ubuntu-bionic https://zuul.opendev.org/t/openstack/build/1aa9f8e1e40a423ab25e470b2b378a5c : SUCCESS in 1h 25m 59s\n- openstack-tox-docs https://zuul.opendev.org/t/openstack/build/70f2c0c051754fdb9a2bd29115e80469 : SUCCESS in 7m 01s\n- build-openstack-releasenotes https://zuul.opendev.org/t/openstack/build/255227ab426149f7bbaf2653e80366b3 : SUCCESS in 5m 43s","accounts_in_message":[],"_revision_number":2},{"id":"3662d0530528cdafffc9bed000a65fa87da04676","author":{"_account_id":28619,"name":"Dmitriy Rabotyagov","email":"noonedeadpunk@gmail.com","username":"noonedeadpunk"},"date":"2020-04-02 19:09:04.000000000","message":"Patch Set 2:\n\nrecheck","accounts_in_message":[],"_revision_number":2},{"id":"10d33540c1b83617acfcdd54dc5ff491f972de55","tag":"autogenerated:zuul:check","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2020-04-02 22:42:57.000000000","message":"Patch Set 2:\n\nBuild failed (check pipeline).  For information on how to proceed, see\nhttps://docs.opendev.org/opendev/infra-manual/latest/developers.html#automated-testing\n\n\n- openstack-ansible-linters https://zuul.opendev.org/t/openstack/build/f10a51fd4be944f0a6aa062724070637 : SUCCESS in 7m 22s\n- openstack-ansible-deploy-aio_metal-debian-buster https://zuul.opendev.org/t/openstack/build/4e35d3400fbc46b4ae348d5e70b419fb : FAILURE in 1h 00m 12s\n- openstack-ansible-deploy-aio_metal-debian-stable https://zuul.opendev.org/t/openstack/build/854eb4c886f64f11bf25183334bf856f : FAILURE in 1h 07m 50s\n- openstack-ansible-deploy-aio_metal-centos-7 https://zuul.opendev.org/t/openstack/build/de60cb3f34ac4f83967b05fd47840396 : FAILURE in 1h 28m 38s\n- openstack-ansible-deploy-aio_metal-opensuse-15 https://zuul.opendev.org/t/openstack/build/8dd9b07e771d4447bc0bba9cf3873c93 : FAILURE in 1h 20m 58s\n- openstack-ansible-deploy-aio_metal-ubuntu-bionic https://zuul.opendev.org/t/openstack/build/f0f0a90851a04aceb26226a80e425c47 : FAILURE in 1h 03m 22s\n- openstack-ansible-upgrade-aio_metal-centos-7 https://zuul.opendev.org/t/openstack/build/e6fa09dd8f2b4dfead55b71b3307f942 : FAILURE in 1h 10m 08s (non-voting)\n- openstack-ansible-upgrade-aio_metal-ubuntu-bionic https://zuul.opendev.org/t/openstack/build/46fee29374864aab84598ac944c7ac76 : FAILURE in 1h 47m 06s\n- openstack-ansible-deploy-aio_distro_metal-centos-7 https://zuul.opendev.org/t/openstack/build/52f173b8eae546248abf33031f83b0cc : SUCCESS in 1h 25m 49s\n- openstack-ansible-deploy-aio_distro_metal-opensuse-15 https://zuul.opendev.org/t/openstack/build/236ce6fe5f3140f0bbd781eb801f652c : FAILURE in 31m 50s (non-voting)\n- openstack-ansible-deploy-aio_distro_metal-ubuntu-bionic https://zuul.opendev.org/t/openstack/build/57e57130178e40bfba109adba333632b : SUCCESS in 1h 04m 40s\n- openstack-ansible-deploy-aio_telemetry_metal-centos-7 https://zuul.opendev.org/t/openstack/build/d334e93dfcb0407388cc3eea52125b2f : FAILURE in 1h 20m 29s\n- openstack-ansible-deploy-aio_telemetry_metal-debian-buster https://zuul.opendev.org/t/openstack/build/fc83d936a6cf4b8ea9dbfe1bc3ac25d5 : SUCCESS in 1h 30m 46s\n- openstack-ansible-deploy-aio_telemetry_metal-debian-stable https://zuul.opendev.org/t/openstack/build/333da8277e694f34a0cd86700365fe77 : FAILURE in 1h 15m 45s\n- openstack-ansible-deploy-aio_telemetry_metal-ubuntu-bionic https://zuul.opendev.org/t/openstack/build/d58f141473b04100918c6f351bdb5e91 : SUCCESS in 1h 25m 23s\n- openstack-tox-docs https://zuul.opendev.org/t/openstack/build/8708ac21a2384dd4b7aa8520ad3b08cb : SUCCESS in 7m 51s\n- build-openstack-releasenotes https://zuul.opendev.org/t/openstack/build/554a1960bf4b4b63adc6699da6c2af31 : SUCCESS in 5m 41s","accounts_in_message":[],"_revision_number":2},{"id":"cf19c9e6948472bae4dd5f75c690d8edb88bdcdc","author":{"_account_id":28619,"name":"Dmitriy Rabotyagov","email":"noonedeadpunk@gmail.com","username":"noonedeadpunk"},"date":"2020-04-06 13:30:14.000000000","message":"Patch Set 2:\n\nrecheck","accounts_in_message":[],"_revision_number":2},{"id":"2d0cc523f0efd4b2ec3508281805b628d1ad0963","tag":"autogenerated:zuul:check","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2020-04-06 16:40:04.000000000","message":"Patch Set 2: Verified+1\n\nBuild succeeded (check pipeline).\n\n- openstack-ansible-linters https://zuul.opendev.org/t/openstack/build/bde23a9b906648d1b885c0d040f9ec2c : SUCCESS in 11m 15s\n- openstack-ansible-deploy-aio_metal-debian-buster https://zuul.opendev.org/t/openstack/build/f14787d934a74dc581155bb03ab54b81 : SUCCESS in 1h 33m 13s\n- openstack-ansible-deploy-aio_metal-debian-stable https://zuul.opendev.org/t/openstack/build/a6c4ee2365a84deeb19060351a424c1b : SUCCESS in 1h 58m 08s\n- openstack-ansible-deploy-aio_metal-centos-7 https://zuul.opendev.org/t/openstack/build/a9b1740d30be4c63a8d9971a64152eb7 : SUCCESS in 2h 04m 31s\n- openstack-ansible-deploy-aio_metal-opensuse-15 https://zuul.opendev.org/t/openstack/build/840493a2f96449e29c46913fba86fc10 : SUCCESS in 1h 40m 45s\n- openstack-ansible-deploy-aio_metal-ubuntu-bionic https://zuul.opendev.org/t/openstack/build/331e6bf9c0304aa6b097b7432632ed22 : SUCCESS in 1h 32m 46s\n- openstack-ansible-upgrade-aio_metal-centos-7 https://zuul.opendev.org/t/openstack/build/d16b8f74048b40d5952913540b07a3f9 : TIMED_OUT in 3h 03m 40s (non-voting)\n- openstack-ansible-upgrade-aio_metal-ubuntu-bionic https://zuul.opendev.org/t/openstack/build/85d5c58fa91b42e691cebfc5701e445f : SUCCESS in 2h 22m 01s\n- openstack-ansible-deploy-aio_distro_metal-centos-7 https://zuul.opendev.org/t/openstack/build/df9d31b633624899a92d2be2fec95b5b : SUCCESS in 1h 39m 55s\n- openstack-ansible-deploy-aio_distro_metal-opensuse-15 https://zuul.opendev.org/t/openstack/build/2af591ca7904405fb73fd45e00f691eb : FAILURE in 31m 30s (non-voting)\n- openstack-ansible-deploy-aio_distro_metal-ubuntu-bionic https://zuul.opendev.org/t/openstack/build/af261d9906c04561a3c31bef1a3a1980 : SUCCESS in 1h 05m 52s\n- openstack-ansible-deploy-aio_telemetry_metal-centos-7 https://zuul.opendev.org/t/openstack/build/07d428d6ec7046d1b977016d270744f3 : SUCCESS in 2h 01m 46s\n- openstack-ansible-deploy-aio_telemetry_metal-debian-buster https://zuul.opendev.org/t/openstack/build/a8eb05e85dfd4edd906958477839a191 : SUCCESS in 1h 33m 13s\n- openstack-ansible-deploy-aio_telemetry_metal-debian-stable https://zuul.opendev.org/t/openstack/build/3930a553ab9348debc13261d9bbfe483 : SUCCESS in 1h 39m 30s\n- openstack-ansible-deploy-aio_telemetry_metal-ubuntu-bionic https://zuul.opendev.org/t/openstack/build/7114726f4d234f4fa3a2ac866b97dc87 : SUCCESS in 1h 32m 33s\n- openstack-tox-docs https://zuul.opendev.org/t/openstack/build/36346adbf25848d683e914644aaf2b23 : SUCCESS in 5m 39s\n- build-openstack-releasenotes https://zuul.opendev.org/t/openstack/build/b0f112726e624c3db803e3240f9d5e24 : SUCCESS in 6m 28s","accounts_in_message":[],"_revision_number":2},{"id":"1ab6fa5679ee07abe8f430160db20b068333e6bf","tag":"autogenerated:zuul:gate","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2020-04-06 16:40:17.000000000","message":"Patch Set 2: -Verified\n\nStarting gate jobs.","accounts_in_message":[],"_revision_number":2},{"id":"9eb67c99037509cfb9207865f2336998c63b6f14","tag":"autogenerated:zuul:gate","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2020-04-06 19:12:40.000000000","message":"Patch Set 2: Verified+2\n\nBuild succeeded (gate pipeline).\n\n- openstack-ansible-linters https://zuul.opendev.org/t/openstack/build/808ad8c86b6b49cabbb9092083d74d34 : SUCCESS in 8m 43s\n- openstack-ansible-deploy-aio_metal-debian-buster https://zuul.opendev.org/t/openstack/build/9c9b875279834eb486e6af5f5cf2220e : SUCCESS in 1h 24m 31s\n- openstack-ansible-deploy-aio_metal-debian-stable https://zuul.opendev.org/t/openstack/build/40fa9a30fc4248ccb0973c88b431e4f6 : SUCCESS in 1h 44m 18s\n- openstack-ansible-deploy-aio_metal-centos-7 https://zuul.opendev.org/t/openstack/build/93ed1454673f49dc988887d5a0063050 : SUCCESS in 2h 09m 39s\n- openstack-ansible-deploy-aio_metal-opensuse-15 https://zuul.opendev.org/t/openstack/build/f0ddc75e660a436ca7d05f39639e7ec7 : SUCCESS in 1h 43m 34s\n- openstack-ansible-deploy-aio_metal-ubuntu-bionic https://zuul.opendev.org/t/openstack/build/d197fb50dc024c1bbb79fc6ef3597b29 : SUCCESS in 1h 38m 16s\n- openstack-ansible-upgrade-aio_metal-ubuntu-bionic https://zuul.opendev.org/t/openstack/build/3d0e605acbb244288f7774a1f9770e06 : SUCCESS in 2h 24m 55s\n- openstack-ansible-deploy-aio_distro_metal-centos-7 https://zuul.opendev.org/t/openstack/build/ea0040e4256448358e7cc40c95f40537 : SUCCESS in 1h 22m 23s\n- openstack-ansible-deploy-aio_distro_metal-ubuntu-bionic https://zuul.opendev.org/t/openstack/build/12ff636145ef430aa686c7842a852571 : SUCCESS in 1h 11m 29s\n- openstack-ansible-deploy-aio_telemetry_metal-centos-7 https://zuul.opendev.org/t/openstack/build/4c7695309b0b49ffacdd192b95294040 : SUCCESS in 1h 49m 57s\n- openstack-ansible-deploy-aio_telemetry_metal-debian-buster https://zuul.opendev.org/t/openstack/build/96e4f7ca7c874343a5f80cf901661ccd : SUCCESS in 1h 19m 04s\n- openstack-ansible-deploy-aio_telemetry_metal-debian-stable https://zuul.opendev.org/t/openstack/build/69af8d483b84490db412eb0bb9a8cbbb : SUCCESS in 1h 54m 25s\n- openstack-ansible-deploy-aio_telemetry_metal-ubuntu-bionic https://zuul.opendev.org/t/openstack/build/472fb75e442947dc90dc830d1930b48c : SUCCESS in 1h 24m 27s\n- openstack-tox-docs https://zuul.opendev.org/t/openstack/build/a7f8910d22724a8ab8eafd354467c0f4 : SUCCESS in 8m 19s\n- build-openstack-releasenotes https://zuul.opendev.org/t/openstack/build/86ec06a9cff34d97ac2236a5ca9b8812 : SUCCESS in 7m 07s","accounts_in_message":[],"_revision_number":2},{"id":"6d48923c1eacceaecf12cbb0fd739e9d4142bfce","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2020-04-06 19:12:41.000000000","message":"Change has been successfully merged by Zuul","accounts_in_message":[],"_revision_number":2},{"id":"7a1a0c1fd3679f56e0d73d78f4e22522d8c4c1e1","tag":"autogenerated:zuul:promote","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2020-04-06 19:14:20.000000000","message":"Patch Set 2:\n\nBuild succeeded (promote pipeline).\n\n- promote-openstack-tox-docs https://zuul.opendev.org/t/openstack/build/07b45a85f6764bb8a7eb1444de4e77e3 : SUCCESS in 1m 21s\n- promote-openstack-releasenotes https://zuul.opendev.org/t/openstack/build/fcd6990e59114e15b7dab01a0454e960 : SUCCESS in 1m 18s","accounts_in_message":[],"_revision_number":2}],"current_revision_number":2,"current_revision":"e140cbb9c74b761b47cce597f5b825d6fcde619c","revisions":{"a9c4f52ecb07d1baf25ac4e49ce3c4d41689e6fc":{"kind":"REWORK","_number":1,"created":"2019-10-22 19:15:55.000000000","uploader":{"_account_id":11131,"name":"OpenStack Proposal Bot","username":"proposal-bot","tags":["SERVICE_USER"]},"ref":"refs/changes/46/690346/1","fetch":{"anonymous http":{"url":"https://review.opendev.org/openstack/openstack-ansible-os_aodh","ref":"refs/changes/46/690346/1","commands":{"Checkout":"git fetch https://review.opendev.org/openstack/openstack-ansible-os_aodh refs/changes/46/690346/1 \u0026\u0026 git checkout FETCH_HEAD","Cherry Pick":"git fetch https://review.opendev.org/openstack/openstack-ansible-os_aodh refs/changes/46/690346/1 \u0026\u0026 git cherry-pick FETCH_HEAD","Format Patch":"git fetch https://review.opendev.org/openstack/openstack-ansible-os_aodh refs/changes/46/690346/1 \u0026\u0026 git format-patch -1 --stdout FETCH_HEAD","Pull":"git pull https://review.opendev.org/openstack/openstack-ansible-os_aodh refs/changes/46/690346/1"}}},"commit":{"parents":[{"commit":"b633e2127bf3e548a73155e7d3598adb52f60aa4","subject":"PDF Documentation Build tox target","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/openstack-ansible-os_aodh/commit/b633e2127bf3e548a73155e7d3598adb52f60aa4"}]}],"author":{"name":"OpenStack Proposal Bot","email":"openstack-infra@lists.openstack.org","date":"2019-10-22 19:15:51.000000000","tz":0},"committer":{"name":"OpenStack Proposal Bot","email":"openstack-infra@lists.openstack.org","date":"2019-10-22 19:15:51.000000000","tz":0},"subject":"Updated from OpenStack Ansible Tests","message":"Updated from OpenStack Ansible Tests\n\nChange-Id: I53997de1aab611a3627a861fdebacaf66c39688b\n","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/openstack-ansible-os_aodh/commit/a9c4f52ecb07d1baf25ac4e49ce3c4d41689e6fc"}],"resolve_conflicts_web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/openstack-ansible-os_aodh/commit/a9c4f52ecb07d1baf25ac4e49ce3c4d41689e6fc"}]},"branch":"refs/heads/stable/train"},"e140cbb9c74b761b47cce597f5b825d6fcde619c":{"kind":"REWORK","_number":2,"created":"2020-03-24 18:33:30.000000000","uploader":{"_account_id":11131,"name":"OpenStack Proposal Bot","username":"proposal-bot","tags":["SERVICE_USER"]},"ref":"refs/changes/46/690346/2","fetch":{"anonymous http":{"url":"https://review.opendev.org/openstack/openstack-ansible-os_aodh","ref":"refs/changes/46/690346/2","commands":{"Checkout":"git fetch https://review.opendev.org/openstack/openstack-ansible-os_aodh refs/changes/46/690346/2 \u0026\u0026 git checkout FETCH_HEAD","Cherry Pick":"git fetch https://review.opendev.org/openstack/openstack-ansible-os_aodh refs/changes/46/690346/2 \u0026\u0026 git cherry-pick FETCH_HEAD","Format Patch":"git fetch https://review.opendev.org/openstack/openstack-ansible-os_aodh refs/changes/46/690346/2 \u0026\u0026 git format-patch -1 --stdout FETCH_HEAD","Pull":"git pull https://review.opendev.org/openstack/openstack-ansible-os_aodh refs/changes/46/690346/2"}}},"commit":{"parents":[{"commit":"cf60654911ef2e55e869f06b4b0630b5d1a9acd2","subject":"Update TOX/UPPER_CONSTRAINTS_FILE for stable/train","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/openstack-ansible-os_aodh/commit/cf60654911ef2e55e869f06b4b0630b5d1a9acd2"}]}],"author":{"name":"OpenStack Proposal Bot","email":"openstack-infra@lists.openstack.org","date":"2020-03-24 18:33:28.000000000","tz":0},"committer":{"name":"OpenStack Proposal Bot","email":"openstack-infra@lists.openstack.org","date":"2020-03-24 18:33:28.000000000","tz":0},"subject":"Updated from OpenStack Ansible Tests","message":"Updated from OpenStack Ansible Tests\n\nChange-Id: I53997de1aab611a3627a861fdebacaf66c39688b\n","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/openstack-ansible-os_aodh/commit/e140cbb9c74b761b47cce597f5b825d6fcde619c"}],"resolve_conflicts_web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/openstack-ansible-os_aodh/commit/e140cbb9c74b761b47cce597f5b825d6fcde619c"}]},"branch":"refs/heads/stable/train"}},"requirements":[],"submit_records":[],"submit_requirements":[]}
