)]}'
{"id":"openstack%2Fopenstack-ansible-os_aodh~709479","triplet_id":"openstack%2Fopenstack-ansible-os_aodh~master~Iba11d032980abeeaa93b2a6f8c046c48e11f5d24","project":"openstack/openstack-ansible-os_aodh","branch":"master","topic":"openstack/openstack-ansible-tests/sync-tests","hashtags":[],"change_id":"Iba11d032980abeeaa93b2a6f8c046c48e11f5d24","subject":"Updated from OpenStack Ansible Tests","status":"MERGED","created":"2020-02-24 13:08:05.000000000","updated":"2020-04-16 13:00:31.000000000","submitted":"2020-04-16 12:57:15.000000000","submitter":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"total_comment_count":0,"unresolved_comment_count":0,"has_review_started":true,"submission_id":"709479-1587041835020-cc994f1b","meta_rev_id":"99143703879ad0e49c968e02cdc14b6275106acc","_number":709479,"virtual_id_number":709479,"owner":{"_account_id":11131,"name":"OpenStack Proposal Bot","username":"proposal-bot","tags":["SERVICE_USER"]},"actions":{},"labels":{"Verified":{"approved":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"all":[{"value":0,"date":"2020-04-14 23:22:40.000000000","_account_id":16011,"name":"James Denton","email":"james.denton@outlook.com","username":"busterswt"},{"tag":"autogenerated:zuul:gate","value":2,"date":"2020-04-16 12:57:14.000000000","permitted_voting_range":{"min":2,"max":2},"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},{"value":0,"_account_id":25023,"name":"Jonathan Rosser","email":"jonathan.rosser@rd.bbc.co.uk","username":"jrosser"}],"values":{"-2":"Fails","-1":"Doesn\u0027t seem to work"," 0":"No score","+1":"Works for me","+2":"Verified"},"description":"","default_value":0,"optional":true},"Code-Review":{"approved":{"_account_id":25023,"name":"Jonathan Rosser","email":"jonathan.rosser@rd.bbc.co.uk","username":"jrosser"},"all":[{"value":0,"_account_id":16011,"name":"James Denton","email":"james.denton@outlook.com","username":"busterswt"},{"value":0,"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},{"value":2,"date":"2020-04-14 08:40:59.000000000","permitted_voting_range":{"min":2,"max":2},"_account_id":25023,"name":"Jonathan Rosser","email":"jonathan.rosser@rd.bbc.co.uk","username":"jrosser"}],"values":{"-2":"Do not merge","-1":"This patch needs further work before it can be merged"," 0":"No score","+1":"Looks good to me, but someone else must approve","+2":"Looks good to me (core reviewer)"},"description":"","default_value":0,"optional":true},"Workflow":{"approved":{"_account_id":25023,"name":"Jonathan Rosser","email":"jonathan.rosser@rd.bbc.co.uk","username":"jrosser"},"all":[{"value":0,"_account_id":16011,"name":"James Denton","email":"james.denton@outlook.com","username":"busterswt"},{"value":0,"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},{"value":1,"date":"2020-04-14 08:40:59.000000000","permitted_voting_range":{"min":1,"max":1},"_account_id":25023,"name":"Jonathan Rosser","email":"jonathan.rosser@rd.bbc.co.uk","username":"jrosser"}],"values":{"-1":"Work in progress"," 0":"Ready for reviews","+1":"Approved"},"description":"","default_value":0,"optional":true},"Backport-Candidate":{"all":[{"value":0,"_account_id":16011,"name":"James Denton","email":"james.denton@outlook.com","username":"busterswt"},{"value":0,"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},{"value":0,"_account_id":25023,"name":"Jonathan Rosser","email":"jonathan.rosser@rd.bbc.co.uk","username":"jrosser"}],"values":{"-1":"Not A Backport Candidate"," 0":"Backport Review Needed","+1":"Proposed Backport"},"description":"","default_value":0,"optional":true}},"removable_reviewers":[],"reviewers":{"REVIEWER":[{"_account_id":16011,"name":"James Denton","email":"james.denton@outlook.com","username":"busterswt"},{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},{"_account_id":25023,"name":"Jonathan Rosser","email":"jonathan.rosser@rd.bbc.co.uk","username":"jrosser"}]},"pending_reviewers":{},"reviewer_updates":[{"updated":"2020-04-14 08:40:59.000000000","updated_by":{"_account_id":25023,"name":"Jonathan Rosser","email":"jonathan.rosser@rd.bbc.co.uk","username":"jrosser"},"reviewer":{"_account_id":25023,"name":"Jonathan Rosser","email":"jonathan.rosser@rd.bbc.co.uk","username":"jrosser"},"state":"REVIEWER"},{"updated":"2020-04-14 23:22:40.000000000","updated_by":{"_account_id":16011,"name":"James Denton","email":"james.denton@outlook.com","username":"busterswt"},"reviewer":{"_account_id":16011,"name":"James Denton","email":"james.denton@outlook.com","username":"busterswt"},"state":"REVIEWER"},{"updated":"2020-04-16 12:57:14.000000000","updated_by":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"reviewer":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"state":"REVIEWER"}],"messages":[{"id":"f54acc125d526cbd5c36dc80ce312a8b73c0efc9","author":{"_account_id":11131,"name":"OpenStack Proposal Bot","username":"proposal-bot","tags":["SERVICE_USER"]},"date":"2020-02-24 13:08:05.000000000","message":"Uploaded patch set 1.","accounts_in_message":[],"_revision_number":1},{"id":"443fd3d14e69ee8c5a1e4a28d750bda7e16f1c66","tag":"autogenerated:zuul:check","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2020-02-24 15:13:26.000000000","message":"Patch Set 1: Verified-1\n\nBuild failed (check pipeline).  For information on how to proceed, see\nhttp://docs.openstack.org/infra/manual/developers.html#automated-testing\n\n\n- openstack-ansible-linters https://zuul.opendev.org/t/openstack/build/1448acee298e48d99d10b528f8ed18fc : SUCCESS in 7m 44s\n- openstack-ansible-deploy-aio_metal-debian-buster https://zuul.opendev.org/t/openstack/build/7f478183cfbb498ba741dac14c9cf708 : SUCCESS in 1h 34m 50s\n- openstack-ansible-deploy-aio_metal-centos-7 https://zuul.opendev.org/t/openstack/build/04af2bd948aa4c088465382f226034f8 : FAILURE in 1h 25m 24s\n- openstack-ansible-deploy-aio_metal-opensuse-15 https://zuul.opendev.org/t/openstack/build/42e1366fa60643c194e368e75515b8b7 : SUCCESS in 1h 40m 36s\n- openstack-ansible-deploy-aio_metal-ubuntu-bionic https://zuul.opendev.org/t/openstack/build/44b568da98734a50b3568cc860a20bea : SUCCESS in 1h 35m 02s\n- openstack-ansible-upgrade-aio_metal-centos-7 https://zuul.opendev.org/t/openstack/build/2fd12700f66b43ba859b199bd2494d9b : FAILURE in 1h 53m 31s (non-voting)\n- openstack-ansible-upgrade-aio_metal-ubuntu-bionic https://zuul.opendev.org/t/openstack/build/b89d24e3d65246a092717788c17e9383 : FAILURE in 1h 17m 28s\n- openstack-ansible-deploy-aio_distro_metal-centos-7 https://zuul.opendev.org/t/openstack/build/9b88d35eff524a9e941f0bb69faab56c : SUCCESS in 1h 59m 00s\n- openstack-ansible-deploy-aio_distro_metal-opensuse-15 https://zuul.opendev.org/t/openstack/build/7338a95612e444d08c7d7b541e46f4de : FAILURE in 32m 12s (non-voting)\n- openstack-ansible-deploy-aio_distro_metal-ubuntu-bionic https://zuul.opendev.org/t/openstack/build/6817e956cedb465280faf8df2b45dbca : SUCCESS in 1h 12m 32s\n- openstack-ansible-deploy-aio_telemetry_metal-centos-7 https://zuul.opendev.org/t/openstack/build/4e82a87f3c96444a889c179f327a6ecd : FAILURE in 1h 28m 59s\n- openstack-ansible-deploy-aio_telemetry_metal-debian-buster https://zuul.opendev.org/t/openstack/build/08e3a0cf25a54517bafa8dc7ced960a6 : SUCCESS in 1h 12m 33s\n- openstack-ansible-deploy-aio_telemetry_metal-ubuntu-bionic https://zuul.opendev.org/t/openstack/build/87d89bb049f7411e8f589e012373a9ee : SUCCESS in 1h 32m 05s\n- openstack-tox-docs https://zuul.opendev.org/t/openstack/build/6c923617ddb74825ba4aa34f7651d92e : SUCCESS in 11m 33s\n- build-openstack-releasenotes https://zuul.opendev.org/t/openstack/build/2805175d7aa041e894d5b3f61ac5f2e9 : SUCCESS in 6m 51s","accounts_in_message":[],"_revision_number":1},{"id":"261440166966194256213a99eac867f46b3ed270","author":{"_account_id":25023,"name":"Jonathan Rosser","email":"jonathan.rosser@rd.bbc.co.uk","username":"jrosser"},"date":"2020-02-25 08:50:25.000000000","message":"Patch Set 1:\n\nrecheck","accounts_in_message":[],"_revision_number":1},{"id":"a5e46bf117a0fabbd76da39617797c3b9761472b","tag":"autogenerated:zuul:check","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2020-02-25 10:48:38.000000000","message":"Patch Set 1:\n\nBuild failed (check pipeline).  For information on how to proceed, see\nhttp://docs.openstack.org/infra/manual/developers.html#automated-testing\n\n\n- openstack-ansible-linters https://zuul.opendev.org/t/openstack/build/61b76d6c328740708cd963af1641de4a : SUCCESS in 9m 47s\n- openstack-ansible-deploy-aio_metal-debian-buster https://zuul.opendev.org/t/openstack/build/721bc10b97674d90b22e27143cd05d00 : SUCCESS in 1h 44m 19s\n- openstack-ansible-deploy-aio_metal-centos-7 https://zuul.opendev.org/t/openstack/build/970257d3b2cc422e8cbc7305d9edbe0e : FAILURE in 1h 31m 17s\n- openstack-ansible-deploy-aio_metal-opensuse-15 https://zuul.opendev.org/t/openstack/build/88276ea3cb794c5eadbd38200e54a005 : FAILURE in 1h 20m 37s\n- openstack-ansible-deploy-aio_metal-ubuntu-bionic https://zuul.opendev.org/t/openstack/build/a445c4a1a4094499bbafd19aa04d8a50 : SUCCESS in 1h 25m 47s\n- openstack-ansible-upgrade-aio_metal-centos-7 https://zuul.opendev.org/t/openstack/build/d44246ccd6fb48589c0618a0e7c6932b : FAILURE in 1h 40m 37s (non-voting)\n- openstack-ansible-upgrade-aio_metal-ubuntu-bionic https://zuul.opendev.org/t/openstack/build/f01a0ec1173e4f7d936b5607bbe1d66f : FAILURE in 1h 28m 17s (non-voting)\n- openstack-ansible-deploy-aio_distro_metal-centos-7 https://zuul.opendev.org/t/openstack/build/2451cb0cbf4948ab844fa8161e68e1cf : SUCCESS in 1h 44m 40s\n- openstack-ansible-deploy-aio_distro_metal-opensuse-15 https://zuul.opendev.org/t/openstack/build/1d9d9f9692f74cff85e2da42a10fb484 : FAILURE in 32m 00s (non-voting)\n- openstack-ansible-deploy-aio_distro_metal-ubuntu-bionic https://zuul.opendev.org/t/openstack/build/33bc35d25acd45af8d03e3542f2d8c49 : SUCCESS in 1h 12m 50s\n- openstack-ansible-deploy-aio_telemetry_metal-centos-7 https://zuul.opendev.org/t/openstack/build/4a04ccaeb6ac4c9a8b7e08269bd4b21e : FAILURE in 1h 09m 44s\n- openstack-ansible-deploy-aio_telemetry_metal-debian-buster https://zuul.opendev.org/t/openstack/build/334edb355d7944c59dc3afefd39579d1 : SUCCESS in 1h 18m 59s\n- openstack-ansible-deploy-aio_telemetry_metal-ubuntu-bionic https://zuul.opendev.org/t/openstack/build/98fadeafeb0f4cc180dddfce02576c94 : FAILURE in 1h 22m 20s\n- openstack-tox-docs https://zuul.opendev.org/t/openstack/build/c5cd809f832f476292f62c29e3ff3d8d : SUCCESS in 9m 45s\n- build-openstack-releasenotes https://zuul.opendev.org/t/openstack/build/00830ee61c7b46e5a8bdaff3bb08d68f : SUCCESS in 6m 27s","accounts_in_message":[],"_revision_number":1},{"id":"2cd74655ffb76d7fd8c9f1f7e9b4be5dbbc47912","author":{"_account_id":11131,"name":"OpenStack Proposal Bot","username":"proposal-bot","tags":["SERVICE_USER"]},"date":"2020-03-06 17:55:13.000000000","message":"Uploaded patch set 2.","accounts_in_message":[],"_revision_number":2},{"id":"18894a70f16276a6ac8e3e815be5eba6282c40d4","tag":"autogenerated:zuul:check","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2020-03-06 19:57:07.000000000","message":"Patch Set 2: Verified-1\n\nBuild failed (check pipeline).  For information on how to proceed, see\nhttp://docs.openstack.org/infra/manual/developers.html#automated-testing\n\n\n- openstack-ansible-linters https://zuul.opendev.org/t/openstack/build/0e76eef484c94087b002dda6f0afb416 : SUCCESS in 11m 30s\n- openstack-ansible-deploy-aio_metal-debian-buster https://zuul.opendev.org/t/openstack/build/ed38fb77c7bd4de1a186bc5399f505cf : SUCCESS in 1h 30m 46s\n- openstack-ansible-deploy-aio_metal-centos-7 https://zuul.opendev.org/t/openstack/build/01cc80e36b6c4d7c9f2279b0334f7c12 : FAILURE in 1h 33m 15s\n- openstack-ansible-deploy-aio_metal-opensuse-15 https://zuul.opendev.org/t/openstack/build/dba07daf73554108982a9c9971145907 : SUCCESS in 1h 48m 33s\n- openstack-ansible-deploy-aio_metal-ubuntu-bionic https://zuul.opendev.org/t/openstack/build/d3e5fb5870ee41979ecbf52bcc3100de : SUCCESS in 1h 25m 37s\n- openstack-ansible-upgrade-aio_metal-centos-7 https://zuul.opendev.org/t/openstack/build/bf4e8248a6da4719b3c22875f4a18956 : FAILURE in 1h 49m 05s (non-voting)\n- openstack-ansible-upgrade-aio_metal-ubuntu-bionic https://zuul.opendev.org/t/openstack/build/c7bcc080d7a545a0a183974a780ec19f : FAILURE in 1h 13m 48s (non-voting)\n- openstack-ansible-deploy-aio_distro_metal-centos-7 https://zuul.opendev.org/t/openstack/build/a0d6518fa2a74f43804f9da731fabd55 : SUCCESS in 1h 35m 55s\n- openstack-ansible-deploy-aio_distro_metal-opensuse-15 https://zuul.opendev.org/t/openstack/build/c4680503faea41e194f4a6c34ac985d3 : FAILURE in 31m 45s (non-voting)\n- openstack-ansible-deploy-aio_distro_metal-ubuntu-bionic https://zuul.opendev.org/t/openstack/build/ac94095316924eae82420c317f342f54 : SUCCESS in 1h 10m 27s\n- openstack-ansible-deploy-aio_telemetry_metal-centos-7 https://zuul.opendev.org/t/openstack/build/7014e3f7d8ce4fb9866cfd399c1d931e : FAILURE in 1h 20m 52s\n- openstack-ansible-deploy-aio_telemetry_metal-debian-buster https://zuul.opendev.org/t/openstack/build/51e35d18dae440dbb17c909559bcce3c : SUCCESS in 1h 19m 18s\n- openstack-ansible-deploy-aio_telemetry_metal-ubuntu-bionic https://zuul.opendev.org/t/openstack/build/921b23f981a244d4909f54c2d8d5065b : SUCCESS in 1h 27m 55s\n- openstack-tox-docs https://zuul.opendev.org/t/openstack/build/2d6361a8fe3349deaf657e540a820989 : SUCCESS in 5m 13s\n- build-openstack-releasenotes https://zuul.opendev.org/t/openstack/build/f495ad88e8f6454081667ba8dabfbb01 : SUCCESS in 5m 34s","accounts_in_message":[],"_revision_number":2},{"id":"eb294f742a36013bc4899fd3163c17de4992f46d","author":{"_account_id":25023,"name":"Jonathan Rosser","email":"jonathan.rosser@rd.bbc.co.uk","username":"jrosser"},"date":"2020-03-07 19:51:24.000000000","message":"Patch Set 2:\n\nrecheck","accounts_in_message":[],"_revision_number":2},{"id":"82a89789f847476111384caf03a21ad376f498cb","tag":"autogenerated:zuul:check","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2020-03-07 21:41:47.000000000","message":"Patch Set 2:\n\nBuild failed (check pipeline).  For information on how to proceed, see\nhttp://docs.openstack.org/infra/manual/developers.html#automated-testing\n\n\n- openstack-ansible-linters https://zuul.opendev.org/t/openstack/build/dde2d715687f4a4e807d7d3f462b0197 : SUCCESS in 7m 24s\n- openstack-ansible-deploy-aio_metal-debian-buster https://zuul.opendev.org/t/openstack/build/5573ad36f44e40d1bc18a944333626f1 : SUCCESS in 1h 16m 45s\n- openstack-ansible-deploy-aio_metal-centos-7 https://zuul.opendev.org/t/openstack/build/f5b1bde59714490e95469bce5244d6da : FAILURE in 1h 28m 23s\n- openstack-ansible-deploy-aio_metal-opensuse-15 https://zuul.opendev.org/t/openstack/build/20809f490cc144bea8f7e8b4e83f5ea1 : SUCCESS in 1h 38m 51s\n- openstack-ansible-deploy-aio_metal-ubuntu-bionic https://zuul.opendev.org/t/openstack/build/aa7b71f41cdf40128a848710215a9d64 : SUCCESS in 1h 22m 55s\n- openstack-ansible-upgrade-aio_metal-centos-7 https://zuul.opendev.org/t/openstack/build/d64fe7167c714bb1852eaf1ffa25408f : FAILURE in 1h 45m 20s (non-voting)\n- openstack-ansible-upgrade-aio_metal-ubuntu-bionic https://zuul.opendev.org/t/openstack/build/a0822ecf084140cf9b8ffcdadf8d7014 : FAILURE in 1h 12m 08s (non-voting)\n- openstack-ansible-deploy-aio_distro_metal-centos-7 https://zuul.opendev.org/t/openstack/build/7080e4fa5c9548498398e863d0863abe : SUCCESS in 1h 32m 15s\n- openstack-ansible-deploy-aio_distro_metal-opensuse-15 https://zuul.opendev.org/t/openstack/build/264dacc0ac4c4c99903c830b5147dff9 : FAILURE in 27m 54s (non-voting)\n- openstack-ansible-deploy-aio_distro_metal-ubuntu-bionic https://zuul.opendev.org/t/openstack/build/c2022b0f0fb4404bb825b2e96b39ace1 : SUCCESS in 1h 08m 29s\n- openstack-ansible-deploy-aio_telemetry_metal-centos-7 https://zuul.opendev.org/t/openstack/build/f38d9acf97aa407eaf919070edb8434d : FAILURE in 1h 21m 09s\n- openstack-ansible-deploy-aio_telemetry_metal-debian-buster https://zuul.opendev.org/t/openstack/build/9ac6eaa94d5a4dc69bb2cdf6869730e5 : SUCCESS in 1h 19m 18s\n- openstack-ansible-deploy-aio_telemetry_metal-ubuntu-bionic https://zuul.opendev.org/t/openstack/build/d2ed959fdbbe4bf79d0b24561d14521a : SUCCESS in 1h 28m 30s\n- openstack-tox-docs https://zuul.opendev.org/t/openstack/build/f6941a17f83d448daa0b9065d89c77f2 : SUCCESS in 5m 18s\n- build-openstack-releasenotes https://zuul.opendev.org/t/openstack/build/52c51d8934e245fe8ce944a8944fa3a8 : SUCCESS in 3m 56s","accounts_in_message":[],"_revision_number":2},{"id":"8f32351408cb10e53466db24d2cfb6f2b9cc215c","author":{"_account_id":11131,"name":"OpenStack Proposal Bot","username":"proposal-bot","tags":["SERVICE_USER"]},"date":"2020-03-11 12:45:54.000000000","message":"Uploaded patch set 3.","accounts_in_message":[],"_revision_number":3},{"id":"9124533151dc3af3d5c1e68bcb551d04bc0c9a63","tag":"autogenerated:zuul:check","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2020-03-11 15:45:26.000000000","message":"Patch Set 3: Verified-1\n\nBuild failed (check pipeline).  For information on how to proceed, see\nhttp://docs.openstack.org/infra/manual/developers.html#automated-testing\n\n\n- openstack-ansible-linters https://zuul.opendev.org/t/openstack/build/0155a3d373b3462caf60bb3553cdeeb0 : SUCCESS in 8m 00s\n- openstack-ansible-deploy-aio_metal-debian-buster https://zuul.opendev.org/t/openstack/build/7c7c7440cd1543db8220b55f3baa9391 : SUCCESS in 1h 22m 46s\n- openstack-ansible-deploy-aio_metal-centos-7 https://zuul.opendev.org/t/openstack/build/e8faf40a4a804d40ac0ec3fa71dc03f8 : FAILURE in 1h 30m 44s\n- openstack-ansible-deploy-aio_metal-opensuse-15 https://zuul.opendev.org/t/openstack/build/f6da877a11794adf91eb4bce67328f7b : SUCCESS in 2h 00m 15s\n- openstack-ansible-deploy-aio_metal-ubuntu-bionic https://zuul.opendev.org/t/openstack/build/4f02d578f06641809815cb6af771a02b : SUCCESS in 1h 33m 16s\n- openstack-ansible-upgrade-aio_metal-centos-7 https://zuul.opendev.org/t/openstack/build/bde45627ea2047ceba71893cd37b5aa4 : FAILURE in 2h 51m 15s (non-voting)\n- openstack-ansible-upgrade-aio_metal-ubuntu-bionic https://zuul.opendev.org/t/openstack/build/57a2b46b2fda400da66d4b20faa99d30 : SUCCESS in 2h 27m 23s (non-voting)\n- openstack-ansible-deploy-aio_distro_metal-centos-7 https://zuul.opendev.org/t/openstack/build/2b68a34dda054919ab463306a519dcf8 : SUCCESS in 1h 36m 33s\n- openstack-ansible-deploy-aio_distro_metal-opensuse-15 https://zuul.opendev.org/t/openstack/build/13970d93c9d24a8db5ac0ac964f118ae : FAILURE in 34m 59s (non-voting)\n- openstack-ansible-deploy-aio_distro_metal-ubuntu-bionic https://zuul.opendev.org/t/openstack/build/d80b2d00311c4650b08ff46c9e79403e : SUCCESS in 1h 29m 50s\n- openstack-ansible-deploy-aio_telemetry_metal-centos-7 https://zuul.opendev.org/t/openstack/build/a4f81bddcaa140ca9c02e1a39b7a0c3d : FAILURE in 1h 36m 56s\n- openstack-ansible-deploy-aio_telemetry_metal-debian-buster https://zuul.opendev.org/t/openstack/build/add8082ef5ea4b4eb997813155fa2c7a : SUCCESS in 1h 19m 13s\n- openstack-ansible-deploy-aio_telemetry_metal-ubuntu-bionic https://zuul.opendev.org/t/openstack/build/f3ac30b536b946b3af2113d18f39fbe3 : SUCCESS in 1h 40m 09s\n- openstack-tox-docs https://zuul.opendev.org/t/openstack/build/8dcdb81af67a45e7891808c288d83a58 : SUCCESS in 6m 54s\n- build-openstack-releasenotes https://zuul.opendev.org/t/openstack/build/1781acfeedc34e95ab03e2697e44532a : SUCCESS in 4m 47s","accounts_in_message":[],"_revision_number":3},{"id":"0d2c631d86baa4e429896c1d01bffb956ae034ef","author":{"_account_id":11131,"name":"OpenStack Proposal Bot","username":"proposal-bot","tags":["SERVICE_USER"]},"date":"2020-04-13 17:47:04.000000000","message":"Uploaded patch set 4.","accounts_in_message":[],"_revision_number":4},{"id":"d5d4ef2bba227065c22f1da6ca5376f0a4496dca","tag":"autogenerated:zuul:check","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2020-04-13 20:58:40.000000000","message":"Patch Set 4: Verified+1\n\nBuild succeeded (check pipeline).\n\n- requirements-check https://zuul.opendev.org/t/openstack/build/b76abef2fbf34e24b2121c4521df206b : SUCCESS in 3m 11s\n- openstack-ansible-linters https://zuul.opendev.org/t/openstack/build/f7548eaf9a644782a80c90d282edd3c4 : SUCCESS in 8m 12s\n- openstack-ansible-deploy-aio_metal-debian-buster https://zuul.opendev.org/t/openstack/build/63db0589ce2d41c0bc5a7c56318fbd55 : SUCCESS in 1h 26m 36s\n- openstack-ansible-deploy-aio_metal-centos-7 https://zuul.opendev.org/t/openstack/build/7c93db035cf34b0aa40b8e9eb7e7edeb : SUCCESS in 2h 08m 23s\n- openstack-ansible-deploy-aio_metal-opensuse-15 https://zuul.opendev.org/t/openstack/build/83134347601d49ad8d0e8b3be04f0d03 : SUCCESS in 1h 50m 42s\n- openstack-ansible-deploy-aio_metal-ubuntu-bionic https://zuul.opendev.org/t/openstack/build/2ec18a1e342840c5927d060f77f1d636 : SUCCESS in 1h 35m 06s\n- openstack-ansible-upgrade-aio_metal-centos-7 https://zuul.opendev.org/t/openstack/build/fc9eb584c9264554b78a3b2fa1cb2d48 : TIMED_OUT in 3h 03m 32s (non-voting)\n- openstack-ansible-upgrade-aio_metal-ubuntu-bionic https://zuul.opendev.org/t/openstack/build/3a2895ce0a4e40bab3402a694c2e28fc : SUCCESS in 2h 19m 44s\n- openstack-ansible-deploy-aio_distro_metal-centos-7 https://zuul.opendev.org/t/openstack/build/3ce6a9fa1ce244a1b6c285f7e15801ba : SUCCESS in 1h 33m 57s\n- openstack-ansible-deploy-aio_distro_metal-opensuse-15 https://zuul.opendev.org/t/openstack/build/ca7845f0d00d44f1a7c1210095dcc1a1 : FAILURE in 41m 14s (non-voting)\n- openstack-ansible-deploy-aio_distro_metal-ubuntu-bionic https://zuul.opendev.org/t/openstack/build/aa83d787a5e24faabc9122a804d79276 : SUCCESS in 1h 10m 34s\n- openstack-ansible-deploy-aio_telemetry_metal-centos-7 https://zuul.opendev.org/t/openstack/build/68213575f18547dbba2c3af71e4e9bfb : SUCCESS in 2h 16m 27s\n- openstack-ansible-deploy-aio_telemetry_metal-debian-buster https://zuul.opendev.org/t/openstack/build/e6372c23c973413cb80c6f1e7f1a341e : SUCCESS in 1h 28m 08s\n- openstack-ansible-deploy-aio_telemetry_metal-ubuntu-bionic https://zuul.opendev.org/t/openstack/build/3dac61b759b44012a8fbde728465dfa5 : SUCCESS in 1h 34m 30s\n- openstack-tox-docs https://zuul.opendev.org/t/openstack/build/1d74a6e0e77a4f92baa2092dc79e19f7 : SUCCESS in 7m 04s\n- build-openstack-releasenotes https://zuul.opendev.org/t/openstack/build/5047406dc1e746db991e99c2fb39d9db : SUCCESS in 4m 32s","accounts_in_message":[],"_revision_number":4},{"id":"1d52018b4713c3cc7846983508a0eef6c2b05961","author":{"_account_id":25023,"name":"Jonathan Rosser","email":"jonathan.rosser@rd.bbc.co.uk","username":"jrosser"},"date":"2020-04-14 08:40:59.000000000","message":"Patch Set 4: Code-Review+2 Workflow+1","accounts_in_message":[],"_revision_number":4},{"id":"66bc3f0cc0d98f8fb1330370668d15d20e42d444","tag":"autogenerated:zuul:gate","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2020-04-14 08:41:11.000000000","message":"Patch Set 4: -Verified\n\nStarting gate jobs.","accounts_in_message":[],"_revision_number":4},{"id":"6144a7125641ce43213a639ca42e0fff6500caed","tag":"autogenerated:zuul:gate","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2020-04-14 11:03:48.000000000","message":"Patch Set 4: Verified-2\n\nBuild failed (gate pipeline).  For information on how to proceed, see\nhttps://docs.opendev.org/opendev/infra-manual/latest/developers.html#automated-testing\n\n\n- requirements-check https://zuul.opendev.org/t/openstack/build/d666e7a5fe6b4a62b3ab61fa555d1762 : SUCCESS in 3m 15s\n- openstack-ansible-linters https://zuul.opendev.org/t/openstack/build/729c1525506a44fbb8ccaf82007da54d : SUCCESS in 6m 59s\n- openstack-ansible-deploy-aio_metal-debian-buster https://zuul.opendev.org/t/openstack/build/6a528fd4d68b460bb2b2d33ef23b8159 : FAILURE in 1h 09m 44s\n- openstack-ansible-deploy-aio_metal-centos-7 https://zuul.opendev.org/t/openstack/build/73c87180f1794a39a060b7d8a1d9ca1e : SUCCESS in 2h 08m 49s\n- openstack-ansible-deploy-aio_metal-opensuse-15 https://zuul.opendev.org/t/openstack/build/d43c5e19ce6e4dc183b153383dfb14d9 : SUCCESS in 1h 43m 31s\n- openstack-ansible-deploy-aio_metal-ubuntu-bionic https://zuul.opendev.org/t/openstack/build/c53e5a5b7f9d4d4f8b72abb772f92b53 : SUCCESS in 1h 44m 49s\n- openstack-ansible-deploy-aio_distro_metal-centos-7 https://zuul.opendev.org/t/openstack/build/28fe1f90fed54ee5952b3c3342236fb4 : SUCCESS in 1h 37m 49s\n- openstack-ansible-deploy-aio_distro_metal-ubuntu-bionic https://zuul.opendev.org/t/openstack/build/ef4700929092430d93cb12cf50523d8f : SUCCESS in 1h 20m 11s\n- openstack-ansible-deploy-aio_telemetry_metal-centos-7 https://zuul.opendev.org/t/openstack/build/b00a148a931445ddac4236df9c667c46 : SUCCESS in 2h 01m 57s\n- openstack-ansible-deploy-aio_telemetry_metal-debian-buster https://zuul.opendev.org/t/openstack/build/f635232888c74bf5860305a76abcaef3 : SUCCESS in 1h 19m 47s\n- openstack-ansible-deploy-aio_telemetry_metal-ubuntu-bionic https://zuul.opendev.org/t/openstack/build/a9f468958f89411986fb753453ec1519 : SUCCESS in 1h 28m 24s\n- openstack-tox-docs https://zuul.opendev.org/t/openstack/build/d509311e190f4a7ab065865f4b72294c : SUCCESS in 5m 55s\n- build-openstack-releasenotes https://zuul.opendev.org/t/openstack/build/7b1928914b0a4de7bb5b75b5c74ae4f4 : SUCCESS in 5m 00s","accounts_in_message":[],"_revision_number":4},{"id":"19ee5c908d8aa489f982b58bfe56ddbc6b6acd48","author":{"_account_id":25023,"name":"Jonathan Rosser","email":"jonathan.rosser@rd.bbc.co.uk","username":"jrosser"},"date":"2020-04-14 17:27:02.000000000","message":"Patch Set 4:\n\nrecheck","accounts_in_message":[],"_revision_number":4},{"id":"920ccff2f960362839319284bf9489d812520ba4","tag":"autogenerated:zuul:check","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2020-04-14 20:40:11.000000000","message":"Patch Set 4: Verified+1\n\nBuild succeeded (check pipeline).\n\n- requirements-check https://zuul.opendev.org/t/openstack/build/308c9a740a1a4cc68dcad5b47237d7d5 : SUCCESS in 2m 36s\n- openstack-ansible-linters https://zuul.opendev.org/t/openstack/build/928c0e778032445798143364770ef637 : SUCCESS in 6m 52s\n- openstack-ansible-deploy-aio_metal-debian-buster https://zuul.opendev.org/t/openstack/build/d52f916b95794bd49c0cbc9c77df03af : SUCCESS in 1h 26m 15s\n- openstack-ansible-deploy-aio_metal-centos-7 https://zuul.opendev.org/t/openstack/build/5855e81e9bf3493ebeda98cee22cefea : SUCCESS in 2h 20m 55s\n- openstack-ansible-deploy-aio_metal-opensuse-15 https://zuul.opendev.org/t/openstack/build/027d5fd6118e4a7c9e5e98c704d9b708 : SUCCESS in 1h 52m 05s\n- openstack-ansible-deploy-aio_metal-ubuntu-bionic https://zuul.opendev.org/t/openstack/build/88b8895b66ac46268366b691703c912c : SUCCESS in 1h 34m 32s\n- openstack-ansible-upgrade-aio_metal-centos-7 https://zuul.opendev.org/t/openstack/build/f4b9db653999434db455602469dcc5b7 : TIMED_OUT in 3h 03m 46s (non-voting)\n- openstack-ansible-upgrade-aio_metal-ubuntu-bionic https://zuul.opendev.org/t/openstack/build/d9f5b7f4b91244fc983e9d33fb2497ba : SUCCESS in 2h 52m 26s\n- openstack-ansible-deploy-aio_distro_metal-centos-7 https://zuul.opendev.org/t/openstack/build/5f485f267c4840d5a494c8b2126a159c : SUCCESS in 1h 23m 36s\n- openstack-ansible-deploy-aio_distro_metal-opensuse-15 https://zuul.opendev.org/t/openstack/build/7a6660cd026c46878c666efc3ce58c07 : FAILURE in 39m 00s (non-voting)\n- openstack-ansible-deploy-aio_distro_metal-ubuntu-bionic https://zuul.opendev.org/t/openstack/build/7100b3023bf44d96ba8706bb8ebfb7ba : SUCCESS in 1h 10m 34s\n- openstack-ansible-deploy-aio_telemetry_metal-centos-7 https://zuul.opendev.org/t/openstack/build/8de85ae3fe3b4bc9b3ca196a8851a020 : SUCCESS in 2h 15m 35s\n- openstack-ansible-deploy-aio_telemetry_metal-debian-buster https://zuul.opendev.org/t/openstack/build/49d17341a9204675aed24a383b8dc86c : SUCCESS in 1h 18m 29s\n- openstack-ansible-deploy-aio_telemetry_metal-ubuntu-bionic https://zuul.opendev.org/t/openstack/build/4739a5cda06a44a2905aa16efb90c8e9 : SUCCESS in 1h 28m 59s\n- openstack-tox-docs https://zuul.opendev.org/t/openstack/build/f940e0654a9e4da1b575f9d3359d93f7 : SUCCESS in 6m 36s\n- build-openstack-releasenotes https://zuul.opendev.org/t/openstack/build/6c4ea0e7135244dda8250258a65e4852 : SUCCESS in 5m 09s","accounts_in_message":[],"_revision_number":4},{"id":"a0fbf01553674297c39cd032815e004b3a63284a","tag":"autogenerated:zuul:gate","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2020-04-14 20:40:23.000000000","message":"Patch Set 4: -Verified\n\nStarting gate jobs.","accounts_in_message":[],"_revision_number":4},{"id":"f943986ea8e6e2f307383ee400aad0142c392071","tag":"autogenerated:zuul:gate","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2020-04-14 22:35:17.000000000","message":"Patch Set 4: Verified-2\n\nBuild failed (gate pipeline).  For information on how to proceed, see\nhttps://docs.opendev.org/opendev/infra-manual/latest/developers.html#automated-testing\n\n\n- requirements-check https://zuul.opendev.org/t/openstack/build/7ef97074e68645ee95eeae23a6743b9d : SUCCESS in 2m 57s\n- openstack-ansible-linters https://zuul.opendev.org/t/openstack/build/41831a552b8c4c5ca953f094b2ca17c4 : SUCCESS in 9m 16s\n- openstack-ansible-deploy-aio_metal-debian-buster https://zuul.opendev.org/t/openstack/build/325660e835084a15a187b00fb8ad5e10 : FAILURE in 58m 20s\n- openstack-ansible-deploy-aio_metal-centos-7 https://zuul.opendev.org/t/openstack/build/d3b52a8236ad4af39010364ab7835c33 : SUCCESS in 1h 48m 21s\n- openstack-ansible-deploy-aio_metal-opensuse-15 https://zuul.opendev.org/t/openstack/build/81f9eb63954a4d49a24d1eda8e1dbd23 : SUCCESS in 1h 45m 46s\n- openstack-ansible-deploy-aio_metal-ubuntu-bionic https://zuul.opendev.org/t/openstack/build/1d35d062a0ee43cab32a09d2ffc0f6f2 : SUCCESS in 1h 23m 34s\n- openstack-ansible-deploy-aio_distro_metal-centos-7 https://zuul.opendev.org/t/openstack/build/69a6edc0bd9a4a51a6cee375c6976285 : SUCCESS in 1h 19m 54s\n- openstack-ansible-deploy-aio_distro_metal-ubuntu-bionic https://zuul.opendev.org/t/openstack/build/3f5988296573471f8f80d52d71cbe5ef : SUCCESS in 1h 29m 17s\n- openstack-ansible-deploy-aio_telemetry_metal-centos-7 https://zuul.opendev.org/t/openstack/build/a65543dc4816425aa822cefec3a4fca7 : SUCCESS in 1h 49m 12s\n- openstack-ansible-deploy-aio_telemetry_metal-debian-buster https://zuul.opendev.org/t/openstack/build/c695da12397d4487a5bf9ea310e70efd : SUCCESS in 1h 20m 00s\n- openstack-ansible-deploy-aio_telemetry_metal-ubuntu-bionic https://zuul.opendev.org/t/openstack/build/7bcde0a5e0a647bfbd5f7dd9f2742c8c : SUCCESS in 1h 30m 45s\n- openstack-tox-docs https://zuul.opendev.org/t/openstack/build/4ace82098ee3439db24ba2e172d809bf : SUCCESS in 6m 10s\n- build-openstack-releasenotes https://zuul.opendev.org/t/openstack/build/2b7b4d1ab6c94ef8a29647c843a92f72 : SUCCESS in 4m 33s","accounts_in_message":[],"_revision_number":4},{"id":"8f072b24d9149758eab15a6b5d77806cc3bd55c2","author":{"_account_id":16011,"name":"James Denton","email":"james.denton@outlook.com","username":"busterswt"},"date":"2020-04-14 23:22:40.000000000","message":"Patch Set 4:\n\nrecheck","accounts_in_message":[],"_revision_number":4},{"id":"f3e2bb00b6b6233b10b8268f5af1112e12b01e42","tag":"autogenerated:zuul:check","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2020-04-15 02:37:06.000000000","message":"Patch Set 4: Verified-1\n\nBuild failed (check pipeline).  For information on how to proceed, see\nhttps://docs.opendev.org/opendev/infra-manual/latest/developers.html#automated-testing\n\n\n- requirements-check https://zuul.opendev.org/t/openstack/build/4f9e9ffee7574b74991c59fc6157a8d2 : SUCCESS in 3m 50s\n- openstack-ansible-linters https://zuul.opendev.org/t/openstack/build/ce04926002c54cc0a2fbffd369ed2d6d : SUCCESS in 7m 43s\n- openstack-ansible-deploy-aio_metal-debian-buster https://zuul.opendev.org/t/openstack/build/8d4db40037dc45e9a77d9db35c120beb : FAILURE in 1h 09m 27s\n- openstack-ansible-deploy-aio_metal-centos-7 https://zuul.opendev.org/t/openstack/build/3ed1df15d30443c89a766e7b21259737 : SUCCESS in 1h 50m 16s\n- openstack-ansible-deploy-aio_metal-opensuse-15 https://zuul.opendev.org/t/openstack/build/15a53967f88e49a18fe94d4a853d5a3e : SUCCESS in 1h 41m 08s\n- openstack-ansible-deploy-aio_metal-ubuntu-bionic https://zuul.opendev.org/t/openstack/build/5e9c7900ac3b4ed698881c9e2efb6ffe : SUCCESS in 1h 31m 01s\n- openstack-ansible-upgrade-aio_metal-centos-7 https://zuul.opendev.org/t/openstack/build/c9fb2e2b0140495cadb97a59b6ab58eb : TIMED_OUT in 3h 04m 20s (non-voting)\n- openstack-ansible-upgrade-aio_metal-ubuntu-bionic https://zuul.opendev.org/t/openstack/build/627746065a474bd2b895e51845b75799 : SUCCESS in 2h 34m 24s\n- openstack-ansible-deploy-aio_distro_metal-centos-7 https://zuul.opendev.org/t/openstack/build/69bb0ea2ac8349ad9eaf1a478f0681b9 : SUCCESS in 1h 30m 00s\n- openstack-ansible-deploy-aio_distro_metal-opensuse-15 https://zuul.opendev.org/t/openstack/build/bc4b29899fe546c59bf720eed98d715f : FAILURE in 35m 20s (non-voting)\n- openstack-ansible-deploy-aio_distro_metal-ubuntu-bionic https://zuul.opendev.org/t/openstack/build/c8d92f5e825b43a2be815bd7dfd3fc2a : SUCCESS in 1h 10m 29s\n- openstack-ansible-deploy-aio_telemetry_metal-centos-7 https://zuul.opendev.org/t/openstack/build/78915a3fdedc47a49448f16106330e7c : SUCCESS in 2h 01m 25s\n- openstack-ansible-deploy-aio_telemetry_metal-debian-buster https://zuul.opendev.org/t/openstack/build/df148313310144bbb16be693c550aee2 : SUCCESS in 1h 26m 49s\n- openstack-ansible-deploy-aio_telemetry_metal-ubuntu-bionic https://zuul.opendev.org/t/openstack/build/098a8251c1404d40b287c55b18c7c42c : SUCCESS in 1h 48m 41s\n- openstack-tox-docs https://zuul.opendev.org/t/openstack/build/244d3d04819448d09b3c9b6994bf7c28 : SUCCESS in 7m 02s\n- build-openstack-releasenotes https://zuul.opendev.org/t/openstack/build/b96f63fc180943f6b7b9c4c8ab51d11f : SUCCESS in 5m 58s","accounts_in_message":[],"_revision_number":4},{"id":"ef605c889f057131851c89be732d8e1e822abd38","author":{"_account_id":25023,"name":"Jonathan Rosser","email":"jonathan.rosser@rd.bbc.co.uk","username":"jrosser"},"date":"2020-04-16 07:27:34.000000000","message":"Patch Set 4:\n\nrecheck","accounts_in_message":[],"_revision_number":4},{"id":"43380388200a39d80ffa2eaf00e65c934311ff85","tag":"autogenerated:zuul:check","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2020-04-16 10:46:53.000000000","message":"Patch Set 4: Verified+1\n\nBuild succeeded (check pipeline).\n\n- requirements-check https://zuul.opendev.org/t/openstack/build/729c3a2cae4c4588919b91a6c5f9aead : SUCCESS in 3m 40s\n- openstack-ansible-linters https://zuul.opendev.org/t/openstack/build/bee8e87cc8d74bee9e1b95c877685ea9 : SUCCESS in 7m 52s\n- openstack-ansible-deploy-aio_metal-debian-buster https://zuul.opendev.org/t/openstack/build/0b8635d70d7b446c8bcb677f40a82172 : SUCCESS in 1h 16m 35s\n- openstack-ansible-deploy-aio_metal-centos-7 https://zuul.opendev.org/t/openstack/build/931be269e4124866bf2f61ba5ddbbba1 : SUCCESS in 2h 06m 29s\n- openstack-ansible-deploy-aio_metal-opensuse-15 https://zuul.opendev.org/t/openstack/build/f15209122f8b4987ae0a83c6987af970 : SUCCESS in 1h 41m 50s\n- openstack-ansible-deploy-aio_metal-ubuntu-bionic https://zuul.opendev.org/t/openstack/build/d51c306cdf60495fa1c8c55be5189b7a : SUCCESS in 1h 28m 27s\n- openstack-ansible-upgrade-aio_metal-centos-7 https://zuul.opendev.org/t/openstack/build/e2318517eb2f4c789e449c37502c504f : TIMED_OUT in 3h 04m 31s (non-voting)\n- openstack-ansible-upgrade-aio_metal-ubuntu-bionic https://zuul.opendev.org/t/openstack/build/de3b712e29c84c41a95ae36ebf008c28 : SUCCESS in 2h 41m 41s\n- openstack-ansible-deploy-aio_distro_metal-centos-7 https://zuul.opendev.org/t/openstack/build/60679826115741fd8dd20544ce70f720 : SUCCESS in 1h 31m 33s\n- openstack-ansible-deploy-aio_distro_metal-opensuse-15 https://zuul.opendev.org/t/openstack/build/f5591c5c86e144038158db9670751c4c : FAILURE in 37m 15s (non-voting)\n- openstack-ansible-deploy-aio_distro_metal-ubuntu-bionic https://zuul.opendev.org/t/openstack/build/f11812493f284ed0b8da379d2d7270f8 : SUCCESS in 1h 13m 02s\n- openstack-ansible-deploy-aio_telemetry_metal-centos-7 https://zuul.opendev.org/t/openstack/build/fd33c0a960424e229fcdc6a963f3f43f : SUCCESS in 2h 06m 14s\n- openstack-ansible-deploy-aio_telemetry_metal-debian-buster https://zuul.opendev.org/t/openstack/build/832258fa1dcc4482be7b6c2c25968292 : SUCCESS in 1h 21m 47s\n- openstack-ansible-deploy-aio_telemetry_metal-ubuntu-bionic https://zuul.opendev.org/t/openstack/build/74f804664578488d85da947d8cf77ac4 : SUCCESS in 1h 38m 25s\n- openstack-tox-docs https://zuul.opendev.org/t/openstack/build/78c08ab35a8a446c94d72e17ed10391a : SUCCESS in 6m 08s\n- build-openstack-releasenotes https://zuul.opendev.org/t/openstack/build/488be440d1bf432aa1aa4d8e633dd218 : SUCCESS in 5m 33s","accounts_in_message":[],"_revision_number":4},{"id":"4bae128ef8709f0bc6dcdec3963192680c8c4de4","tag":"autogenerated:zuul:gate","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2020-04-16 10:47:06.000000000","message":"Patch Set 4: -Verified\n\nStarting gate jobs.","accounts_in_message":[],"_revision_number":4},{"id":"6d1035c4117cb7074ae5a0a0b9a210a0af5ff1b1","tag":"autogenerated:zuul:gate","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2020-04-16 12:57:14.000000000","message":"Patch Set 4: Verified+2\n\nBuild succeeded (gate pipeline).\n\n- requirements-check https://zuul.opendev.org/t/openstack/build/7301bb5eba7a486cbdc334984db47758 : SUCCESS in 2m 57s\n- openstack-ansible-linters https://zuul.opendev.org/t/openstack/build/910ad0e561994c59a15bf2548f64ebb0 : SUCCESS in 6m 03s\n- openstack-ansible-deploy-aio_metal-debian-buster https://zuul.opendev.org/t/openstack/build/d0c304de63fc4c06ac22579e2146153d : SUCCESS in 1h 10m 39s\n- openstack-ansible-deploy-aio_metal-centos-7 https://zuul.opendev.org/t/openstack/build/306d8e9d3db54e648b4d3893006395ee : SUCCESS in 2h 04m 43s\n- openstack-ansible-deploy-aio_metal-opensuse-15 https://zuul.opendev.org/t/openstack/build/c923d8086aed4e71b15ec26ef8b57291 : SUCCESS in 1h 44m 49s\n- openstack-ansible-deploy-aio_metal-ubuntu-bionic https://zuul.opendev.org/t/openstack/build/839dc48b14fa4c5496e32dbd8756217e : SUCCESS in 1h 47m 49s\n- openstack-ansible-deploy-aio_distro_metal-centos-7 https://zuul.opendev.org/t/openstack/build/0d97a221c9914942bffc848925638042 : SUCCESS in 1h 35m 21s\n- openstack-ansible-deploy-aio_distro_metal-ubuntu-bionic https://zuul.opendev.org/t/openstack/build/7f161343f162411f8990a9b4d1f5659b : SUCCESS in 1h 25m 35s\n- openstack-ansible-deploy-aio_telemetry_metal-centos-7 https://zuul.opendev.org/t/openstack/build/7adc562045ed47869072e19d40d06bdf : SUCCESS in 1h 36m 42s\n- openstack-ansible-deploy-aio_telemetry_metal-debian-buster https://zuul.opendev.org/t/openstack/build/393a4998d8614084888abad8a8c64626 : SUCCESS in 1h 20m 38s\n- openstack-ansible-deploy-aio_telemetry_metal-ubuntu-bionic https://zuul.opendev.org/t/openstack/build/341038c729f34a5eb28962dd5e93a188 : SUCCESS in 1h 32m 56s\n- openstack-tox-docs https://zuul.opendev.org/t/openstack/build/bc0c6056dcc54543a3db80aec7508e55 : SUCCESS in 7m 10s\n- build-openstack-releasenotes https://zuul.opendev.org/t/openstack/build/df33d9a0ef054764acd0c9afad249211 : SUCCESS in 4m 03s","accounts_in_message":[],"_revision_number":4},{"id":"5bcc915df26b27387ef7bd510b32e6e2a0311ee8","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2020-04-16 12:57:15.000000000","message":"Change has been successfully merged by Zuul","accounts_in_message":[],"_revision_number":4},{"id":"488749d15f0093147cf43a4d7b2bb7f0702f1de8","tag":"autogenerated:zuul:promote","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2020-04-16 13:00:31.000000000","message":"Patch Set 4:\n\nBuild succeeded (promote pipeline).\n\n- promote-openstack-tox-docs https://zuul.opendev.org/t/openstack/build/3d4a7cd1add7418ca200efa459944e65 : SUCCESS in 2m 43s\n- promote-openstack-releasenotes https://zuul.opendev.org/t/openstack/build/4dd10f488a6c4bccad0841855ea5c9f4 : SUCCESS in 2m 32s","accounts_in_message":[],"_revision_number":4}],"current_revision_number":4,"current_revision":"3875426057c919888abd3b56293e4e1f0b632915","revisions":{"6d44455b2ca8d61788e9d7a78d9d993d7988fc0e":{"kind":"REWORK","_number":1,"created":"2020-02-24 13:08:05.000000000","uploader":{"_account_id":11131,"name":"OpenStack Proposal Bot","username":"proposal-bot","tags":["SERVICE_USER"]},"ref":"refs/changes/79/709479/1","fetch":{"anonymous http":{"url":"https://review.opendev.org/openstack/openstack-ansible-os_aodh","ref":"refs/changes/79/709479/1","commands":{"Checkout":"git fetch https://review.opendev.org/openstack/openstack-ansible-os_aodh refs/changes/79/709479/1 \u0026\u0026 git checkout FETCH_HEAD","Cherry Pick":"git fetch https://review.opendev.org/openstack/openstack-ansible-os_aodh refs/changes/79/709479/1 \u0026\u0026 git cherry-pick FETCH_HEAD","Format Patch":"git fetch https://review.opendev.org/openstack/openstack-ansible-os_aodh refs/changes/79/709479/1 \u0026\u0026 git format-patch -1 --stdout FETCH_HEAD","Pull":"git pull https://review.opendev.org/openstack/openstack-ansible-os_aodh refs/changes/79/709479/1"}}},"commit":{"parents":[{"commit":"3b4c254827598619437bd91f1f3c919dae28a87b","subject":"Merge \"tox: Keeping going with docs\"","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/openstack-ansible-os_aodh/commit/3b4c254827598619437bd91f1f3c919dae28a87b"}]}],"author":{"name":"OpenStack Proposal Bot","email":"openstack-infra@lists.openstack.org","date":"2020-02-24 13:08:01.000000000","tz":0},"committer":{"name":"OpenStack Proposal Bot","email":"openstack-infra@lists.openstack.org","date":"2020-02-24 13:08:01.000000000","tz":0},"subject":"Updated from OpenStack Ansible Tests","message":"Updated from OpenStack Ansible Tests\n\nChange-Id: Iba11d032980abeeaa93b2a6f8c046c48e11f5d24\n","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/openstack-ansible-os_aodh/commit/6d44455b2ca8d61788e9d7a78d9d993d7988fc0e"}],"resolve_conflicts_web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/openstack-ansible-os_aodh/commit/6d44455b2ca8d61788e9d7a78d9d993d7988fc0e"}]},"branch":"refs/heads/master"},"243e7956e4f8058a5c4762581f1533637c7fc9e4":{"kind":"REWORK","_number":2,"created":"2020-03-06 17:55:13.000000000","uploader":{"_account_id":11131,"name":"OpenStack Proposal Bot","username":"proposal-bot","tags":["SERVICE_USER"]},"ref":"refs/changes/79/709479/2","fetch":{"anonymous http":{"url":"https://review.opendev.org/openstack/openstack-ansible-os_aodh","ref":"refs/changes/79/709479/2","commands":{"Checkout":"git fetch https://review.opendev.org/openstack/openstack-ansible-os_aodh refs/changes/79/709479/2 \u0026\u0026 git checkout FETCH_HEAD","Cherry Pick":"git fetch https://review.opendev.org/openstack/openstack-ansible-os_aodh refs/changes/79/709479/2 \u0026\u0026 git cherry-pick FETCH_HEAD","Format Patch":"git fetch https://review.opendev.org/openstack/openstack-ansible-os_aodh refs/changes/79/709479/2 \u0026\u0026 git format-patch -1 --stdout FETCH_HEAD","Pull":"git pull https://review.opendev.org/openstack/openstack-ansible-os_aodh refs/changes/79/709479/2"}}},"commit":{"parents":[{"commit":"e061b4eb949ba9b31631438aee87b0f5149c9f04","subject":"[ussuri][goal] Updates for python 2.7 drop","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/openstack-ansible-os_aodh/commit/e061b4eb949ba9b31631438aee87b0f5149c9f04"}]}],"author":{"name":"OpenStack Proposal Bot","email":"openstack-infra@lists.openstack.org","date":"2020-03-06 17:55:11.000000000","tz":0},"committer":{"name":"OpenStack Proposal Bot","email":"openstack-infra@lists.openstack.org","date":"2020-03-06 17:55:11.000000000","tz":0},"subject":"Updated from OpenStack Ansible Tests","message":"Updated from OpenStack Ansible Tests\n\nChange-Id: Iba11d032980abeeaa93b2a6f8c046c48e11f5d24\n","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/openstack-ansible-os_aodh/commit/243e7956e4f8058a5c4762581f1533637c7fc9e4"}],"resolve_conflicts_web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/openstack-ansible-os_aodh/commit/243e7956e4f8058a5c4762581f1533637c7fc9e4"}]},"branch":"refs/heads/master"},"a2965b26ead3de71b2b6a24592ea0df30d5509fb":{"kind":"REWORK","_number":3,"created":"2020-03-11 12:45:54.000000000","uploader":{"_account_id":11131,"name":"OpenStack Proposal Bot","username":"proposal-bot","tags":["SERVICE_USER"]},"ref":"refs/changes/79/709479/3","fetch":{"anonymous http":{"url":"https://review.opendev.org/openstack/openstack-ansible-os_aodh","ref":"refs/changes/79/709479/3","commands":{"Checkout":"git fetch https://review.opendev.org/openstack/openstack-ansible-os_aodh refs/changes/79/709479/3 \u0026\u0026 git checkout FETCH_HEAD","Cherry Pick":"git fetch https://review.opendev.org/openstack/openstack-ansible-os_aodh refs/changes/79/709479/3 \u0026\u0026 git cherry-pick FETCH_HEAD","Format Patch":"git fetch https://review.opendev.org/openstack/openstack-ansible-os_aodh refs/changes/79/709479/3 \u0026\u0026 git format-patch -1 --stdout FETCH_HEAD","Pull":"git pull https://review.opendev.org/openstack/openstack-ansible-os_aodh refs/changes/79/709479/3"}}},"commit":{"parents":[{"commit":"e061b4eb949ba9b31631438aee87b0f5149c9f04","subject":"[ussuri][goal] Updates for python 2.7 drop","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/openstack-ansible-os_aodh/commit/e061b4eb949ba9b31631438aee87b0f5149c9f04"}]}],"author":{"name":"OpenStack Proposal Bot","email":"openstack-infra@lists.openstack.org","date":"2020-03-11 12:45:53.000000000","tz":0},"committer":{"name":"OpenStack Proposal Bot","email":"openstack-infra@lists.openstack.org","date":"2020-03-11 12:45:53.000000000","tz":0},"subject":"Updated from OpenStack Ansible Tests","message":"Updated from OpenStack Ansible Tests\n\nChange-Id: Iba11d032980abeeaa93b2a6f8c046c48e11f5d24\n","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/openstack-ansible-os_aodh/commit/a2965b26ead3de71b2b6a24592ea0df30d5509fb"}],"resolve_conflicts_web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/openstack-ansible-os_aodh/commit/a2965b26ead3de71b2b6a24592ea0df30d5509fb"}]},"branch":"refs/heads/master"},"3875426057c919888abd3b56293e4e1f0b632915":{"kind":"REWORK","_number":4,"created":"2020-04-13 17:47:04.000000000","uploader":{"_account_id":11131,"name":"OpenStack Proposal Bot","username":"proposal-bot","tags":["SERVICE_USER"]},"ref":"refs/changes/79/709479/4","fetch":{"anonymous http":{"url":"https://review.opendev.org/openstack/openstack-ansible-os_aodh","ref":"refs/changes/79/709479/4","commands":{"Checkout":"git fetch https://review.opendev.org/openstack/openstack-ansible-os_aodh refs/changes/79/709479/4 \u0026\u0026 git checkout FETCH_HEAD","Cherry Pick":"git fetch https://review.opendev.org/openstack/openstack-ansible-os_aodh refs/changes/79/709479/4 \u0026\u0026 git cherry-pick FETCH_HEAD","Format Patch":"git fetch https://review.opendev.org/openstack/openstack-ansible-os_aodh refs/changes/79/709479/4 \u0026\u0026 git format-patch -1 --stdout FETCH_HEAD","Pull":"git pull https://review.opendev.org/openstack/openstack-ansible-os_aodh refs/changes/79/709479/4"}}},"commit":{"parents":[{"commit":"43fd466166b370eb5a9fda7bd2062949bd6d8c97","subject":"Cleanup py27 support","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/openstack-ansible-os_aodh/commit/43fd466166b370eb5a9fda7bd2062949bd6d8c97"}]}],"author":{"name":"OpenStack Proposal Bot","email":"openstack-infra@lists.openstack.org","date":"2020-04-13 17:47:00.000000000","tz":0},"committer":{"name":"OpenStack Proposal Bot","email":"openstack-infra@lists.openstack.org","date":"2020-04-13 17:47:00.000000000","tz":0},"subject":"Updated from OpenStack Ansible Tests","message":"Updated from OpenStack Ansible Tests\n\nChange-Id: Iba11d032980abeeaa93b2a6f8c046c48e11f5d24\n","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/openstack-ansible-os_aodh/commit/3875426057c919888abd3b56293e4e1f0b632915"}],"resolve_conflicts_web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/openstack-ansible-os_aodh/commit/3875426057c919888abd3b56293e4e1f0b632915"}]},"branch":"refs/heads/master"}},"requirements":[],"submit_records":[],"submit_requirements":[]}
