)]}'
{"id":"openstack%2Fopenstack-ansible-os_aodh~712885","triplet_id":"openstack%2Fopenstack-ansible-os_aodh~master~If89c9c608ce82601cb60ae266fe7697a47b7e382","project":"openstack/openstack-ansible-os_aodh","branch":"master","topic":"fix","hashtags":[],"change_id":"If89c9c608ce82601cb60ae266fe7697a47b7e382","subject":"Missing document start \"---\"","status":"ABANDONED","created":"2020-03-13 09:40:27.000000000","updated":"2021-01-19 15:31:11.000000000","total_comment_count":0,"unresolved_comment_count":0,"has_review_started":true,"meta_rev_id":"26887d1f2892ca85cc590281c4b024b29027a6da","_number":712885,"virtual_id_number":712885,"owner":{"_account_id":23317,"name":"ShangXiao","email":"shangxiaobj@inspur.com","username":"ShangXiao"},"actions":{},"labels":{"Verified":{"recommended":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"all":[{"_account_id":6547,"name":"Andreas Jaeger","email":"jaegerandi@gmail.com","username":"jaegerandi"},{"date":"2020-04-13 01:17:52.000000000","_account_id":23317,"name":"ShangXiao","email":"shangxiaobj@inspur.com","username":"ShangXiao"},{"tag":"autogenerated:zuul:check","value":1,"date":"2020-04-13 04:24:06.000000000","permitted_voting_range":{"min":-2,"max":2},"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]}],"values":{"-2":"Fails","-1":"Doesn\u0027t seem to work"," 0":"No score","+1":"Works for me","+2":"Verified"},"description":"","value":1,"default_value":0,"optional":true},"Code-Review":{"disliked":{"_account_id":6547,"name":"Andreas Jaeger","email":"jaegerandi@gmail.com","username":"jaegerandi"},"all":[{"value":-1,"date":"2020-04-13 05:52:12.000000000","permitted_voting_range":{"min":-1,"max":1},"_account_id":6547,"name":"Andreas Jaeger","email":"jaegerandi@gmail.com","username":"jaegerandi"},{"value":0,"permitted_voting_range":{"min":-1,"max":1},"_account_id":23317,"name":"ShangXiao","email":"shangxiaobj@inspur.com","username":"ShangXiao"},{"value":0,"permitted_voting_range":{"min":-1,"max":1},"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]}],"values":{"-2":"Do not merge","-1":"This patch needs further work before it can be merged"," 0":"No score","+1":"Looks good to me, but someone else must approve","+2":"Looks good to me (core reviewer)"},"description":"","value":-1,"default_value":0,"optional":true},"Workflow":{"all":[{"_account_id":6547,"name":"Andreas Jaeger","email":"jaegerandi@gmail.com","username":"jaegerandi"},{"value":0,"permitted_voting_range":{"min":-1,"max":0},"_account_id":23317,"name":"ShangXiao","email":"shangxiaobj@inspur.com","username":"ShangXiao"},{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]}],"values":{"-1":"Work in progress"," 0":"Ready for reviews","+1":"Approved"},"description":"","default_value":0,"optional":true},"Backport-Candidate":{"all":[{"_account_id":6547,"name":"Andreas Jaeger","email":"jaegerandi@gmail.com","username":"jaegerandi"},{"_account_id":23317,"name":"ShangXiao","email":"shangxiaobj@inspur.com","username":"ShangXiao"},{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]}],"values":{"-1":"Not A Backport Candidate"," 0":"Backport Review Needed","+1":"Proposed Backport"},"description":"","default_value":0,"optional":true}},"removable_reviewers":[],"reviewers":{"REVIEWER":[{"_account_id":6547,"name":"Andreas Jaeger","email":"jaegerandi@gmail.com","username":"jaegerandi"},{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},{"_account_id":23317,"name":"ShangXiao","email":"shangxiaobj@inspur.com","username":"ShangXiao"}]},"pending_reviewers":{},"reviewer_updates":[{"updated":"2020-04-13 04:24:06.000000000","updated_by":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"reviewer":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"state":"REVIEWER"},{"updated":"2020-04-13 05:52:12.000000000","updated_by":{"_account_id":6547,"name":"Andreas Jaeger","email":"jaegerandi@gmail.com","username":"jaegerandi"},"reviewer":{"_account_id":6547,"name":"Andreas Jaeger","email":"jaegerandi@gmail.com","username":"jaegerandi"},"state":"REVIEWER"}],"messages":[{"id":"f6de65915f80c250f7a8e33bdb8ee6fccf868c29","author":{"_account_id":23317,"name":"ShangXiao","email":"shangxiaobj@inspur.com","username":"ShangXiao"},"date":"2020-03-13 09:40:27.000000000","message":"Uploaded patch set 1.","accounts_in_message":[],"_revision_number":1},{"id":"2000cb9de562dd1cb4ae7a51fae47b286341faea","tag":"autogenerated:zuul:check","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2020-03-13 12:39:04.000000000","message":"Patch Set 1: Verified-1\n\nBuild failed (check pipeline).  For information on how to proceed, see\nhttp://docs.openstack.org/infra/manual/developers.html#automated-testing\n\n\n- openstack-ansible-linters https://zuul.opendev.org/t/openstack/build/adf45ebc58724335aa4a1e49701c636e : SUCCESS in 6m 48s\n- openstack-ansible-deploy-aio_metal-debian-buster https://zuul.opendev.org/t/openstack/build/c753eb538b10414ab40535a77365ebfd : SUCCESS in 1h 45m 09s\n- openstack-ansible-deploy-aio_metal-centos-7 https://zuul.opendev.org/t/openstack/build/cdb6d18cee154534826dd47e8543fdf3 : FAILURE in 1h 27m 46s\n- openstack-ansible-deploy-aio_metal-opensuse-15 https://zuul.opendev.org/t/openstack/build/9b7abfe99d394214aabf5432aed176ed : SUCCESS in 1h 41m 04s\n- openstack-ansible-deploy-aio_metal-ubuntu-bionic https://zuul.opendev.org/t/openstack/build/0df621a48e224d01b55b485863c3035e : FAILURE in 1h 52m 11s\n- openstack-ansible-upgrade-aio_metal-centos-7 https://zuul.opendev.org/t/openstack/build/635c213594234f06af52d5350afe0631 : FAILURE in 7m 24s (non-voting)\n- openstack-ansible-upgrade-aio_metal-ubuntu-bionic https://zuul.opendev.org/t/openstack/build/10edac177cb44cf7866579a0850a2ee8 : FAILURE in 2h 53m 48s (non-voting)\n- openstack-ansible-deploy-aio_distro_metal-centos-7 https://zuul.opendev.org/t/openstack/build/84420acff5024810b3405e2e1527f5f2 : SUCCESS in 1h 40m 22s\n- openstack-ansible-deploy-aio_distro_metal-opensuse-15 https://zuul.opendev.org/t/openstack/build/95597852147645afb14c68fda2de9886 : FAILURE in 45m 14s (non-voting)\n- openstack-ansible-deploy-aio_distro_metal-ubuntu-bionic https://zuul.opendev.org/t/openstack/build/5f4e5df3fa864044b1ea7e03d4a5d14a : SUCCESS in 1h 15m 23s\n- openstack-ansible-deploy-aio_telemetry_metal-centos-7 https://zuul.opendev.org/t/openstack/build/32d289e1028d4fa1b81de071340b59fb : FAILURE in 1h 41m 35s\n- openstack-ansible-deploy-aio_telemetry_metal-debian-buster https://zuul.opendev.org/t/openstack/build/474455b3b6214000bcb2c00a2a0073da : SUCCESS in 1h 32m 23s\n- openstack-ansible-deploy-aio_telemetry_metal-ubuntu-bionic https://zuul.opendev.org/t/openstack/build/4db1728821ad4269b24f5c8509b72315 : FAILURE in 1h 33m 29s\n- openstack-tox-docs https://zuul.opendev.org/t/openstack/build/03befddb707e4148b96463b55495f3f9 : SUCCESS in 5m 58s","accounts_in_message":[],"_revision_number":1},{"id":"15aa7234380f52e4b1a7e1a40b9e89e1794f50b7","author":{"_account_id":23317,"name":"ShangXiao","email":"shangxiaobj@inspur.com","username":"ShangXiao"},"date":"2020-03-20 00:20:47.000000000","message":"Patch Set 1:\n\nrecheck","accounts_in_message":[],"_revision_number":1},{"id":"1ac5be106fc26d705109ccfe5bc1656e4a461ef6","tag":"autogenerated:zuul:check","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2020-03-20 02:33:39.000000000","message":"Patch Set 1:\n\nBuild failed (check pipeline).  For information on how to proceed, see\nhttp://docs.openstack.org/infra/manual/developers.html#automated-testing\n\n\n- openstack-ansible-linters https://zuul.opendev.org/t/openstack/build/d1198b64e7a349d287ae9dde42a1ac03 : SUCCESS in 6m 01s\n- openstack-ansible-deploy-aio_metal-debian-buster https://zuul.opendev.org/t/openstack/build/e27efbf2910a4eeea32b599aed553217 : SUCCESS in 1h 14m 21s\n- openstack-ansible-deploy-aio_metal-centos-7 https://zuul.opendev.org/t/openstack/build/2d2911c4fbe64ddba47d28e8560c0110 : FAILURE in 1h 19m 56s\n- openstack-ansible-deploy-aio_metal-opensuse-15 https://zuul.opendev.org/t/openstack/build/7c31f2c6f7b94b339ccbb4d2d34794d3 : SUCCESS in 1h 32m 21s\n- openstack-ansible-deploy-aio_metal-ubuntu-bionic https://zuul.opendev.org/t/openstack/build/ea981a59d6f54e82b034c7c7f0f31016 : SUCCESS in 1h 20m 53s\n- openstack-ansible-upgrade-aio_metal-centos-7 https://zuul.opendev.org/t/openstack/build/6319958c30f7402aa1f74a53d019c15d : FAILURE in 7m 17s (non-voting)\n- openstack-ansible-upgrade-aio_metal-ubuntu-bionic https://zuul.opendev.org/t/openstack/build/8b862401455f4f06912039676cb28627 : SUCCESS in 2h 12m 00s (non-voting)\n- openstack-ansible-deploy-aio_distro_metal-centos-7 https://zuul.opendev.org/t/openstack/build/e3e70cb2bf21405ebacfe6a45a7ffae3 : SUCCESS in 1h 25m 01s\n- openstack-ansible-deploy-aio_distro_metal-opensuse-15 https://zuul.opendev.org/t/openstack/build/49909862696349d3af4bcba79c194761 : FAILURE in 31m 07s (non-voting)\n- openstack-ansible-deploy-aio_distro_metal-ubuntu-bionic https://zuul.opendev.org/t/openstack/build/0738d53dbc114c92957a287b0f5d271c : SUCCESS in 1h 04m 00s\n- openstack-ansible-deploy-aio_telemetry_metal-centos-7 https://zuul.opendev.org/t/openstack/build/514fcf85f9e44939ab96b5b224b769bc : FAILURE in 1h 19m 10s\n- openstack-ansible-deploy-aio_telemetry_metal-debian-buster https://zuul.opendev.org/t/openstack/build/b6dfbb88a164424289ae0ef306230a88 : SUCCESS in 1h 12m 59s\n- openstack-ansible-deploy-aio_telemetry_metal-ubuntu-bionic https://zuul.opendev.org/t/openstack/build/3996c0b062024be7b7f0513ae13ca2e6 : SUCCESS in 1h 27m 11s\n- openstack-tox-docs https://zuul.opendev.org/t/openstack/build/9856a79382b546f494d822dec8b7c8d8 : SUCCESS in 4m 15s","accounts_in_message":[],"_revision_number":1},{"id":"0616a10d89165da10db051c84c8235002f5b291e","author":{"_account_id":23317,"name":"ShangXiao","email":"shangxiaobj@inspur.com","username":"ShangXiao"},"date":"2020-03-24 00:33:42.000000000","message":"Patch Set 1:\n\nrecheck","accounts_in_message":[],"_revision_number":1},{"id":"cd72d354cd27a397be7d36891a2d06ff22810954","tag":"autogenerated:zuul:check","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2020-03-24 04:09:28.000000000","message":"Patch Set 1:\n\nBuild failed (check pipeline).  For information on how to proceed, see\nhttps://docs.opendev.org/opendev/infra-manual/latest/developers.html#automated-testing\n\n\n- openstack-ansible-linters https://zuul.opendev.org/t/openstack/build/12e3de7c8c22467f9f1fbc18a1e31001 : SUCCESS in 8m 49s\n- openstack-ansible-deploy-aio_metal-debian-buster https://zuul.opendev.org/t/openstack/build/ad7bc1f8cc6640c09db28656c90a611e : SUCCESS in 1h 22m 35s\n- openstack-ansible-deploy-aio_metal-centos-7 https://zuul.opendev.org/t/openstack/build/29462013fe1140788d7a060c997b26bc : FAILURE in 1h 22m 45s\n- openstack-ansible-deploy-aio_metal-opensuse-15 https://zuul.opendev.org/t/openstack/build/165ca44159864746a43618ea881761f4 : FAILURE in 2h 14m 59s\n- openstack-ansible-deploy-aio_metal-ubuntu-bionic https://zuul.opendev.org/t/openstack/build/ec2306763100423fae018670f844f327 : SUCCESS in 1h 29m 40s\n- openstack-ansible-upgrade-aio_metal-centos-7 https://zuul.opendev.org/t/openstack/build/bc5461f86cac44b2b3fd92285d8d3f89 : FAILURE in 7m 16s (non-voting)\n- openstack-ansible-upgrade-aio_metal-ubuntu-bionic https://zuul.opendev.org/t/openstack/build/0027f510997d406d98c3549a69937ac4 : FAILURE in 2h 43m 13s (non-voting)\n- openstack-ansible-deploy-aio_distro_metal-centos-7 https://zuul.opendev.org/t/openstack/build/22f78bd2c88342e7877b87d24e6b22d6 : TIMED_OUT in 3h 31m 06s\n- openstack-ansible-deploy-aio_distro_metal-opensuse-15 https://zuul.opendev.org/t/openstack/build/f86fe0628fe249e58fb118842b494451 : FAILURE in 33m 11s (non-voting)\n- openstack-ansible-deploy-aio_distro_metal-ubuntu-bionic https://zuul.opendev.org/t/openstack/build/15a43444fca04ad7ac488ace11e3e30e : FAILURE in 2h 05m 58s\n- openstack-ansible-deploy-aio_telemetry_metal-centos-7 https://zuul.opendev.org/t/openstack/build/b4b66be1416b4f64a4d544d425900e19 : FAILURE in 1h 21m 13s\n- openstack-ansible-deploy-aio_telemetry_metal-debian-buster https://zuul.opendev.org/t/openstack/build/749c0f1dde6f40a283b31fa2a23aa38a : FAILURE in 1h 10m 43s\n- openstack-ansible-deploy-aio_telemetry_metal-ubuntu-bionic https://zuul.opendev.org/t/openstack/build/f51c3a1a26fb4752baba4d0d8290a90e : SUCCESS in 1h 27m 25s\n- openstack-tox-docs https://zuul.opendev.org/t/openstack/build/51ca6a55064046019d723909db4b2816 : SUCCESS in 4m 37s","accounts_in_message":[],"_revision_number":1},{"id":"cce41ee75119b3b2adba2295cd964738fecd0f30","author":{"_account_id":23317,"name":"ShangXiao","email":"shangxiaobj@inspur.com","username":"ShangXiao"},"date":"2020-04-13 01:17:52.000000000","message":"Patch Set 1:\n\nrecheck","accounts_in_message":[],"_revision_number":1},{"id":"64889ac8d38a059e1278333bdaa96cc303cff0e6","tag":"autogenerated:zuul:check","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2020-04-13 04:24:06.000000000","message":"Patch Set 1: Verified+1\n\nBuild succeeded (check pipeline).\n\n- openstack-ansible-linters https://zuul.opendev.org/t/openstack/build/3793fe9624aa45a89e3ce210ca0a8b78 : SUCCESS in 5m 35s\n- openstack-ansible-deploy-aio_metal-debian-buster https://zuul.opendev.org/t/openstack/build/09442ace57194efba8f3a1f66654b68d : SUCCESS in 1h 18m 03s\n- openstack-ansible-deploy-aio_metal-centos-7 https://zuul.opendev.org/t/openstack/build/969dddcc22fe4fd58ed08e1d18d7e064 : SUCCESS in 2h 03m 04s\n- openstack-ansible-deploy-aio_metal-opensuse-15 https://zuul.opendev.org/t/openstack/build/001db651da1a4c999a7c1544c4dcab36 : SUCCESS in 1h 44m 15s\n- openstack-ansible-deploy-aio_metal-ubuntu-bionic https://zuul.opendev.org/t/openstack/build/9269f97c24cc4b4a8baf464d9ed9869c : SUCCESS in 1h 27m 32s\n- openstack-ansible-upgrade-aio_metal-centos-7 https://zuul.opendev.org/t/openstack/build/964bd77ab0664b2893ac2599cf146336 : TIMED_OUT in 3h 03m 39s (non-voting)\n- openstack-ansible-upgrade-aio_metal-ubuntu-bionic https://zuul.opendev.org/t/openstack/build/359e28ad3c7b4e5e8a4b96ce3872feaf : SUCCESS in 2h 25m 33s\n- openstack-ansible-deploy-aio_distro_metal-centos-7 https://zuul.opendev.org/t/openstack/build/180e627289ee44f1a78eb371e2cc96fd : SUCCESS in 1h 27m 19s\n- openstack-ansible-deploy-aio_distro_metal-opensuse-15 https://zuul.opendev.org/t/openstack/build/f879c76aa8cd4ab59c9920c0531f0e4f : FAILURE in 36m 58s (non-voting)\n- openstack-ansible-deploy-aio_distro_metal-ubuntu-bionic https://zuul.opendev.org/t/openstack/build/1a56bfbae842450995fc1fc2f7ad7ed5 : SUCCESS in 1h 05m 09s\n- openstack-ansible-deploy-aio_telemetry_metal-centos-7 https://zuul.opendev.org/t/openstack/build/245ba4376651483a972b0c05033fe11c : SUCCESS in 2h 02m 26s\n- openstack-ansible-deploy-aio_telemetry_metal-debian-buster https://zuul.opendev.org/t/openstack/build/cf39315c9f954a65aeee37f83d8bc2db : SUCCESS in 1h 18m 44s\n- openstack-ansible-deploy-aio_telemetry_metal-ubuntu-bionic https://zuul.opendev.org/t/openstack/build/fc703eebe2ce4832863a8d618ff10f2f : SUCCESS in 1h 34m 21s\n- openstack-tox-docs https://zuul.opendev.org/t/openstack/build/34127d9ee1c149d5aefb72bb60fdfffd : SUCCESS in 5m 44s","accounts_in_message":[],"_revision_number":1},{"id":"690236411174cc4d31e05285f43923e0b7272541","author":{"_account_id":6547,"name":"Andreas Jaeger","email":"jaegerandi@gmail.com","username":"jaegerandi"},"date":"2020-04-13 05:52:12.000000000","message":"Patch Set 1: Code-Review-1\n\nthis is not needed for Zuul, works just fine as is.","accounts_in_message":[],"_revision_number":1},{"id":"26887d1f2892ca85cc590281c4b024b29027a6da","tag":"autogenerated:gerrit:abandon","author":{"_account_id":25023,"name":"Jonathan Rosser","email":"jonathan.rosser@rd.bbc.co.uk","username":"jrosser"},"date":"2021-01-19 15:31:11.000000000","message":"Abandoned\n\nthis change is not necessary","accounts_in_message":[],"_revision_number":1}],"current_revision_number":1,"current_revision":"be01563a2e44095e1e0705c966269608bb214f6a","revisions":{"be01563a2e44095e1e0705c966269608bb214f6a":{"kind":"REWORK","_number":1,"created":"2020-03-13 09:40:27.000000000","uploader":{"_account_id":23317,"name":"ShangXiao","email":"shangxiaobj@inspur.com","username":"ShangXiao"},"ref":"refs/changes/85/712885/1","fetch":{"anonymous http":{"url":"https://review.opendev.org/openstack/openstack-ansible-os_aodh","ref":"refs/changes/85/712885/1","commands":{"Checkout":"git fetch https://review.opendev.org/openstack/openstack-ansible-os_aodh refs/changes/85/712885/1 \u0026\u0026 git checkout FETCH_HEAD","Cherry Pick":"git fetch https://review.opendev.org/openstack/openstack-ansible-os_aodh refs/changes/85/712885/1 \u0026\u0026 git cherry-pick FETCH_HEAD","Format Patch":"git fetch https://review.opendev.org/openstack/openstack-ansible-os_aodh refs/changes/85/712885/1 \u0026\u0026 git format-patch -1 --stdout FETCH_HEAD","Pull":"git pull https://review.opendev.org/openstack/openstack-ansible-os_aodh refs/changes/85/712885/1"}}},"commit":{"parents":[{"commit":"e061b4eb949ba9b31631438aee87b0f5149c9f04","subject":"[ussuri][goal] Updates for python 2.7 drop","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/openstack-ansible-os_aodh/commit/e061b4eb949ba9b31631438aee87b0f5149c9f04"}]}],"author":{"name":"ShangXiao","email":"shangxiaobj@inspur.com","date":"2020-03-13 09:39:30.000000000","tz":480},"committer":{"name":"ShangXiao","email":"shangxiaobj@inspur.com","date":"2020-03-13 09:40:17.000000000","tz":480},"subject":"Missing document start \"---\"","message":"Missing document start \"---\"\n\nChange-Id: If89c9c608ce82601cb60ae266fe7697a47b7e382\n","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/openstack-ansible-os_aodh/commit/be01563a2e44095e1e0705c966269608bb214f6a"}],"resolve_conflicts_web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/openstack-ansible-os_aodh/commit/be01563a2e44095e1e0705c966269608bb214f6a"}]},"branch":"refs/heads/master"}},"requirements":[],"submit_records":[],"submit_requirements":[]}
