)]}'
{"id":"openstack%2Fopenstack-ansible-os_aodh~733270","triplet_id":"openstack%2Fopenstack-ansible-os_aodh~master~I9b6949163c74f5248ffea884eb71060c884f8938","project":"openstack/openstack-ansible-os_aodh","branch":"master","topic":"reno-openstackdocstheme","hashtags":[],"change_id":"I9b6949163c74f5248ffea884eb71060c884f8938","subject":"Use newer openstackdocstheme and reno versions","status":"MERGED","created":"2020-06-03 17:11:47.000000000","updated":"2020-06-05 11:42:35.000000000","submitted":"2020-06-05 11:38:42.000000000","submitter":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"total_comment_count":0,"unresolved_comment_count":0,"has_review_started":true,"submission_id":"733270-1591357122684-0e537f8a","meta_rev_id":"80b3861861ae660eb657a5fbb3f564b786b63fec","_number":733270,"virtual_id_number":733270,"owner":{"_account_id":6547,"name":"Andreas Jaeger","email":"jaegerandi@gmail.com","username":"jaegerandi"},"actions":{},"labels":{"Verified":{"approved":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"all":[{"value":0,"_account_id":28619,"name":"Dmitriy Rabotyagov","email":"noonedeadpunk@gmail.com","username":"noonedeadpunk"},{"tag":"autogenerated:zuul:gate","value":2,"date":"2020-06-05 11:38:42.000000000","post_submit":true,"permitted_voting_range":{"min":2,"max":2},"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},{"value":0,"_account_id":25023,"name":"Jonathan Rosser","email":"jonathan.rosser@rd.bbc.co.uk","username":"jrosser"}],"values":{"-2":"Fails","-1":"Doesn\u0027t seem to work"," 0":"No score","+1":"Works for me","+2":"Verified"},"description":"","default_value":0,"optional":true},"Code-Review":{"approved":{"_account_id":28619,"name":"Dmitriy Rabotyagov","email":"noonedeadpunk@gmail.com","username":"noonedeadpunk"},"all":[{"value":2,"date":"2020-06-05 11:25:11.000000000","permitted_voting_range":{"min":2,"max":2},"_account_id":28619,"name":"Dmitriy Rabotyagov","email":"noonedeadpunk@gmail.com","username":"noonedeadpunk"},{"value":0,"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},{"value":2,"date":"2020-06-04 06:11:47.000000000","permitted_voting_range":{"min":2,"max":2},"_account_id":25023,"name":"Jonathan Rosser","email":"jonathan.rosser@rd.bbc.co.uk","username":"jrosser"}],"values":{"-2":"Do not merge","-1":"This patch needs further work before it can be merged"," 0":"No score","+1":"Looks good to me, but someone else must approve","+2":"Looks good to me (core reviewer)"},"description":"","default_value":0,"optional":true},"Workflow":{"approved":{"_account_id":28619,"name":"Dmitriy Rabotyagov","email":"noonedeadpunk@gmail.com","username":"noonedeadpunk"},"all":[{"value":1,"date":"2020-06-05 11:25:11.000000000","permitted_voting_range":{"min":1,"max":1},"_account_id":28619,"name":"Dmitriy Rabotyagov","email":"noonedeadpunk@gmail.com","username":"noonedeadpunk"},{"value":0,"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},{"value":0,"_account_id":25023,"name":"Jonathan Rosser","email":"jonathan.rosser@rd.bbc.co.uk","username":"jrosser"}],"values":{"-1":"Work in progress"," 0":"Ready for reviews","+1":"Approved"},"description":"","default_value":0,"optional":true},"Backport-Candidate":{"all":[{"value":0,"_account_id":28619,"name":"Dmitriy Rabotyagov","email":"noonedeadpunk@gmail.com","username":"noonedeadpunk"},{"value":0,"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},{"value":0,"_account_id":25023,"name":"Jonathan Rosser","email":"jonathan.rosser@rd.bbc.co.uk","username":"jrosser"}],"values":{"-1":"Not A Backport Candidate"," 0":"Backport Review Needed","+1":"Proposed Backport"},"description":"","default_value":0,"optional":true}},"removable_reviewers":[],"reviewers":{"REVIEWER":[{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},{"_account_id":25023,"name":"Jonathan Rosser","email":"jonathan.rosser@rd.bbc.co.uk","username":"jrosser"},{"_account_id":28619,"name":"Dmitriy Rabotyagov","email":"noonedeadpunk@gmail.com","username":"noonedeadpunk"}]},"pending_reviewers":{},"reviewer_updates":[{"updated":"2020-06-04 06:11:47.000000000","updated_by":{"_account_id":25023,"name":"Jonathan Rosser","email":"jonathan.rosser@rd.bbc.co.uk","username":"jrosser"},"reviewer":{"_account_id":25023,"name":"Jonathan Rosser","email":"jonathan.rosser@rd.bbc.co.uk","username":"jrosser"},"state":"REVIEWER"},{"updated":"2020-06-05 11:25:11.000000000","updated_by":{"_account_id":28619,"name":"Dmitriy Rabotyagov","email":"noonedeadpunk@gmail.com","username":"noonedeadpunk"},"reviewer":{"_account_id":28619,"name":"Dmitriy Rabotyagov","email":"noonedeadpunk@gmail.com","username":"noonedeadpunk"},"state":"REVIEWER"},{"updated":"2020-06-05 11:38:42.000000000","updated_by":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"reviewer":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"state":"REVIEWER"}],"messages":[{"id":"a25dc748d9d60943235a70dbcab9a00cea1c511a","author":{"_account_id":6547,"name":"Andreas Jaeger","email":"jaegerandi@gmail.com","username":"jaegerandi"},"date":"2020-06-03 17:11:47.000000000","message":"Uploaded patch set 1.","accounts_in_message":[],"_revision_number":1},{"id":"d7a5a52e30577bb22a1b5a154b625593cb2a52d7","tag":"autogenerated:zuul:check","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2020-06-03 17:23:11.000000000","message":"Patch Set 1: Verified+1\n\nBuild succeeded (check pipeline).\n\n- openstack-ansible-linters https://zuul.opendev.org/t/openstack/build/7d422ba6aa804bfb83818718b7e2ad63 : SUCCESS in 10m 07s\n- openstack-tox-docs https://zuul.opendev.org/t/openstack/build/4c0d8eb4641548bdb8f6a5f53f02100c : SUCCESS in 6m 23s\n- build-openstack-releasenotes https://zuul.opendev.org/t/openstack/build/7f0e3aa44657488db20e6f6cd362b584 : SUCCESS in 4m 06s","accounts_in_message":[],"_revision_number":1},{"id":"1b2c73183d9d586472bec954a4f956ef5511e3b2","author":{"_account_id":25023,"name":"Jonathan Rosser","email":"jonathan.rosser@rd.bbc.co.uk","username":"jrosser"},"date":"2020-06-04 06:11:47.000000000","message":"Patch Set 1: Code-Review+2","accounts_in_message":[],"_revision_number":1},{"id":"38f35ce1178b84de89fa53a2bdd47822f4f81892","author":{"_account_id":28619,"name":"Dmitriy Rabotyagov","email":"noonedeadpunk@gmail.com","username":"noonedeadpunk"},"date":"2020-06-05 11:25:11.000000000","message":"Patch Set 1: Code-Review+2 Workflow+1","accounts_in_message":[],"_revision_number":1},{"id":"c130f39bc63456f1fbe03c8a0a8e592688e3b4ed","tag":"autogenerated:zuul:gate","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2020-06-05 11:25:22.000000000","message":"Patch Set 1: -Verified\n\nStarting gate jobs.","accounts_in_message":[],"_revision_number":1},{"id":"43e5d4a988a20581afc5c059716c2827254f8def","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2020-06-05 11:38:42.000000000","message":"Change has been successfully merged by Zuul","accounts_in_message":[],"_revision_number":1},{"id":"8d60138707219bd7d971ccd9c0f340f2b2d74fff","tag":"autogenerated:zuul:gate","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2020-06-05 11:38:42.000000000","message":"Patch Set 1: Verified+2\n\nBuild succeeded (gate pipeline).\n\n- openstack-ansible-linters https://zuul.opendev.org/t/openstack/build/80122ad305bc44d18758d4ad35915f11 : SUCCESS in 7m 08s\n- openstack-tox-docs https://zuul.opendev.org/t/openstack/build/b5aebbb454d94465bf55e0a8d4932d89 : SUCCESS in 7m 16s\n- build-openstack-releasenotes https://zuul.opendev.org/t/openstack/build/2fa07960c2f44badab645b7a31f59113 : SUCCESS in 3m 48s","accounts_in_message":[],"_revision_number":1},{"id":"0eb21019d9144adf7b7a2a557e1b59a957f589e8","tag":"autogenerated:zuul:promote","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2020-06-05 11:42:35.000000000","message":"Patch Set 1:\n\nBuild succeeded (promote pipeline).\n\n- promote-openstack-tox-docs https://zuul.opendev.org/t/openstack/build/f2484ba053ac4525a8b24e4e1beda742 : SUCCESS in 2m 12s\n- promote-openstack-releasenotes https://zuul.opendev.org/t/openstack/build/0b0ec57a78dd4a798de465d7b7106474 : SUCCESS in 3m 25s","accounts_in_message":[],"_revision_number":1}],"current_revision_number":1,"current_revision":"76f5b5703f22583f9385c6c8a708d409b74d1fc2","revisions":{"76f5b5703f22583f9385c6c8a708d409b74d1fc2":{"kind":"REWORK","_number":1,"created":"2020-06-03 17:11:47.000000000","uploader":{"_account_id":6547,"name":"Andreas Jaeger","email":"jaegerandi@gmail.com","username":"jaegerandi"},"ref":"refs/changes/70/733270/1","fetch":{"anonymous http":{"url":"https://review.opendev.org/openstack/openstack-ansible-os_aodh","ref":"refs/changes/70/733270/1","commands":{"Checkout":"git fetch https://review.opendev.org/openstack/openstack-ansible-os_aodh refs/changes/70/733270/1 \u0026\u0026 git checkout FETCH_HEAD","Cherry Pick":"git fetch https://review.opendev.org/openstack/openstack-ansible-os_aodh refs/changes/70/733270/1 \u0026\u0026 git cherry-pick FETCH_HEAD","Format Patch":"git fetch https://review.opendev.org/openstack/openstack-ansible-os_aodh refs/changes/70/733270/1 \u0026\u0026 git format-patch -1 --stdout FETCH_HEAD","Pull":"git pull https://review.opendev.org/openstack/openstack-ansible-os_aodh refs/changes/70/733270/1"}}},"commit":{"parents":[{"commit":"d65394009765e6e17f25364ee0fb8015d72437d0","subject":"Update master for stable/ussuri","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/openstack-ansible-os_aodh/commit/d65394009765e6e17f25364ee0fb8015d72437d0"}]}],"author":{"name":"Andreas Jaeger","email":"aj@suse.com","date":"2020-06-03 16:27:59.000000000","tz":120},"committer":{"name":"Andreas Jaeger","email":"aj@suse.com","date":"2020-06-03 17:11:39.000000000","tz":120},"subject":"Use newer openstackdocstheme and reno versions","message":"Use newer openstackdocstheme and reno versions\n\nThe sync from https://review.opendev.org/733244 updated to\nopenstackdocstheme 2.2.1 and reno 3.1.0 versions.\n\nSet openstackdocs_pdf_link to link to PDF file. Note that\nthe link to the published document only works on docs.openstack.org\nwhere the PDF file is placed in the top-level html directory. The\nsite-preview places the PDF in a pdf directory.\n\nopenstackdocstheme renames some variables, so follow the renames\nbefore the next release removes them. A couple of variables are also\nnot needed anymore, remove them.\n\nSee also\nhttp://lists.openstack.org/pipermail/openstack-discuss/2020-May/014971.html\n\nChange-Id: I9b6949163c74f5248ffea884eb71060c884f8938\n","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/openstack-ansible-os_aodh/commit/76f5b5703f22583f9385c6c8a708d409b74d1fc2"}],"resolve_conflicts_web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/openstack-ansible-os_aodh/commit/76f5b5703f22583f9385c6c8a708d409b74d1fc2"}]},"branch":"refs/heads/master"}},"requirements":[],"submit_records":[],"submit_requirements":[]}
