)]}'
{"id":"openstack%2Fopenstack-ansible-os_aodh~736206","triplet_id":"openstack%2Fopenstack-ansible-os_aodh~stable%2Fussuri~I98386dde8e10e0ded820326af443aa6e70f53321","project":"openstack/openstack-ansible-os_aodh","branch":"stable/ussuri","topic":"interpreter-stable/ussuri","hashtags":[],"change_id":"I98386dde8e10e0ded820326af443aa6e70f53321","subject":"remove ansible_python_interpreter","status":"MERGED","created":"2020-06-17 13:10:51.000000000","updated":"2020-06-18 08:36:56.000000000","submitted":"2020-06-18 08:34:02.000000000","submitter":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"total_comment_count":0,"unresolved_comment_count":0,"has_review_started":true,"submission_id":"736206-1592469243269-0b4201c0","meta_rev_id":"614c64be1237f5bb0f6d1a0430f258c2173b07e4","_number":736206,"virtual_id_number":736206,"owner":{"_account_id":25023,"name":"Jonathan Rosser","email":"jonathan.rosser@rd.bbc.co.uk","username":"jrosser"},"actions":{},"labels":{"Verified":{"approved":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"all":[{"value":0,"_account_id":13095,"name":"Marc Gariépy","email":"gariepy.marc@gmail.com","username":"mgariepy"},{"tag":"autogenerated:zuul:gate","value":2,"date":"2020-06-18 08:34:02.000000000","permitted_voting_range":{"min":2,"max":2},"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},{"value":0,"_account_id":25023,"name":"Jonathan Rosser","email":"jonathan.rosser@rd.bbc.co.uk","username":"jrosser"}],"values":{"-2":"Fails","-1":"Doesn\u0027t seem to work"," 0":"No score","+1":"Works for me","+2":"Verified"},"description":"","default_value":0,"optional":true},"Code-Review":{"approved":{"_account_id":13095,"name":"Marc Gariépy","email":"gariepy.marc@gmail.com","username":"mgariepy"},"all":[{"value":2,"date":"2020-06-17 20:19:38.000000000","permitted_voting_range":{"min":2,"max":2},"_account_id":13095,"name":"Marc Gariépy","email":"gariepy.marc@gmail.com","username":"mgariepy"},{"value":0,"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},{"value":2,"date":"2020-06-18 06:19:47.000000000","permitted_voting_range":{"min":2,"max":2},"_account_id":25023,"name":"Jonathan Rosser","email":"jonathan.rosser@rd.bbc.co.uk","username":"jrosser"}],"values":{"-2":"Do not merge","-1":"This patch needs further work before it can be merged"," 0":"No score","+1":"Looks good to me, but someone else must approve","+2":"Looks good to me (core reviewer)"},"description":"","default_value":0,"optional":true},"Workflow":{"approved":{"_account_id":25023,"name":"Jonathan Rosser","email":"jonathan.rosser@rd.bbc.co.uk","username":"jrosser"},"all":[{"value":0,"_account_id":13095,"name":"Marc Gariépy","email":"gariepy.marc@gmail.com","username":"mgariepy"},{"value":0,"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},{"value":1,"date":"2020-06-18 06:19:47.000000000","permitted_voting_range":{"min":1,"max":1},"_account_id":25023,"name":"Jonathan Rosser","email":"jonathan.rosser@rd.bbc.co.uk","username":"jrosser"}],"values":{"-1":"Work in progress"," 0":"Ready for reviews","+1":"Approved"},"description":"","default_value":0,"optional":true},"Backport-Candidate":{"all":[{"value":0,"_account_id":13095,"name":"Marc Gariépy","email":"gariepy.marc@gmail.com","username":"mgariepy"},{"value":0,"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},{"value":0,"_account_id":25023,"name":"Jonathan Rosser","email":"jonathan.rosser@rd.bbc.co.uk","username":"jrosser"}],"values":{"-1":"Not A Backport Candidate"," 0":"Backport Review Needed","+1":"Proposed Backport"},"description":"","default_value":0,"optional":true}},"removable_reviewers":[],"reviewers":{"REVIEWER":[{"_account_id":13095,"name":"Marc Gariépy","email":"gariepy.marc@gmail.com","username":"mgariepy"},{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},{"_account_id":25023,"name":"Jonathan Rosser","email":"jonathan.rosser@rd.bbc.co.uk","username":"jrosser"}]},"pending_reviewers":{},"reviewer_updates":[{"updated":"2020-06-17 20:19:38.000000000","updated_by":{"_account_id":13095,"name":"Marc Gariépy","email":"gariepy.marc@gmail.com","username":"mgariepy"},"reviewer":{"_account_id":13095,"name":"Marc Gariépy","email":"gariepy.marc@gmail.com","username":"mgariepy"},"state":"REVIEWER"},{"updated":"2020-06-18 08:34:02.000000000","updated_by":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"reviewer":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"state":"REVIEWER"}],"messages":[{"id":"751722fd86a82e0ae33437afd69f0b0de7852eea","author":{"_account_id":25023,"name":"Jonathan Rosser","email":"jonathan.rosser@rd.bbc.co.uk","username":"jrosser"},"date":"2020-06-17 13:10:51.000000000","message":"Patch Set 1: Cherry Picked from branch master.","accounts_in_message":[],"_revision_number":1},{"id":"e5dfc34c11abe9b921e45af2b28304d3a64cc2a1","tag":"autogenerated:zuul:check","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2020-06-17 16:21:30.000000000","message":"Patch Set 1: Verified+1\n\nBuild succeeded (check pipeline).\n\n- openstack-ansible-linters https://zuul.opendev.org/t/openstack/build/ce594c0cb3714bedb3ee84816925da0b : SUCCESS in 11m 56s\n- openstack-ansible-deploy-aio_metal-debian-buster https://zuul.opendev.org/t/openstack/build/12d4b2065dee4b3ab80c69febfc3e9e5 : SUCCESS in 1h 31m 54s\n- openstack-ansible-deploy-aio_metal-centos-7 https://zuul.opendev.org/t/openstack/build/2bbabcd0aebc44d1ab4301ace1b1b1a4 : SUCCESS in 2h 57m 13s\n- openstack-ansible-deploy-aio_metal-ubuntu-bionic https://zuul.opendev.org/t/openstack/build/68f05a0a1a5a4c43895e1c3708e7f00d : SUCCESS in 1h 27m 49s\n- openstack-ansible-upgrade-aio_metal-centos-7 https://zuul.opendev.org/t/openstack/build/3f8d26de4851456ba5aa886395d68b12 : TIMED_OUT in 3h 04m 41s (non-voting)\n- openstack-ansible-upgrade-aio_metal-ubuntu-bionic https://zuul.opendev.org/t/openstack/build/ca8c8b4f5d5a4c818b5348c80301c6ef : SUCCESS in 2h 23m 12s\n- openstack-ansible-deploy-aio_distro_metal-centos-7 https://zuul.opendev.org/t/openstack/build/8efa93fef10948acb12159632a8a8270 : SUCCESS in 1h 26m 25s\n- openstack-ansible-deploy-aio_distro_metal-ubuntu-bionic https://zuul.opendev.org/t/openstack/build/00417815d43747b981787271517f5357 : SUCCESS in 1h 16m 54s\n- openstack-ansible-deploy-aio_telemetry_metal-centos-7 https://zuul.opendev.org/t/openstack/build/cb9635868ab74d36b099af66a889a3ad : SUCCESS in 1h 16m 58s\n- openstack-ansible-deploy-aio_telemetry_metal-debian-buster https://zuul.opendev.org/t/openstack/build/06f8227066da432d99a10e1f558ecf38 : SUCCESS in 1h 01m 51s\n- openstack-ansible-deploy-aio_telemetry_metal-ubuntu-bionic https://zuul.opendev.org/t/openstack/build/d577c49f096d4dcc88d7ad1eca19541f : SUCCESS in 2h 01m 17s\n- openstack-tox-docs https://zuul.opendev.org/t/openstack/build/6ac43082d44b41d0a0f1f27c047201d0 : SUCCESS in 8m 44s","accounts_in_message":[],"_revision_number":1},{"id":"4ed959f4fe16b84cec9207ffc4ad70ea9081a074","author":{"_account_id":25023,"name":"Jonathan Rosser","email":"jonathan.rosser@rd.bbc.co.uk","username":"jrosser"},"date":"2020-06-17 19:13:17.000000000","message":"Patch Set 1: Code-Review+2","accounts_in_message":[],"_revision_number":1},{"id":"70d342bc405e6a184194dba04ac2a3da2faf3160","author":{"_account_id":13095,"name":"Marc Gariépy","email":"gariepy.marc@gmail.com","username":"mgariepy"},"date":"2020-06-17 19:57:35.000000000","message":"Patch Set 1: Code-Review+2 Workflow+1","accounts_in_message":[],"_revision_number":1},{"id":"38284663352f87b82e7bf657ba30338ec1bc20d9","tag":"autogenerated:zuul:gate","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2020-06-17 19:57:50.000000000","message":"Patch Set 1: -Verified\n\nStarting gate jobs.","accounts_in_message":[],"_revision_number":1},{"id":"9349af45f48f544e64cf2ada39419f682ccefdb7","author":{"_account_id":13095,"name":"Marc Gariépy","email":"gariepy.marc@gmail.com","username":"mgariepy"},"date":"2020-06-17 20:17:01.000000000","message":"Patch Set 2: Commit message was updated.","accounts_in_message":[],"_revision_number":2},{"id":"1d2de29b51296dd9fe0e00733b5e67d45b72c36b","author":{"_account_id":13095,"name":"Marc Gariépy","email":"gariepy.marc@gmail.com","username":"mgariepy"},"date":"2020-06-17 20:19:38.000000000","message":"Patch Set 2: Code-Review+2","accounts_in_message":[],"_revision_number":2},{"id":"394312d34d006b7807bab56244616498cef8ff53","tag":"autogenerated:zuul:check","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2020-06-17 23:56:23.000000000","message":"Patch Set 2: Verified+1\n\nBuild succeeded (check pipeline).\n\n- openstack-ansible-linters https://zuul.opendev.org/t/openstack/build/427ca9441aa34eb3985415eb5fcae3ff : SUCCESS in 8m 05s\n- openstack-ansible-deploy-aio_metal-debian-buster https://zuul.opendev.org/t/openstack/build/a4872ee267fb485b976b82b1993d64fa : SUCCESS in 1h 23m 05s\n- openstack-ansible-deploy-aio_metal-centos-7 https://zuul.opendev.org/t/openstack/build/5991feef96fc42f2b0bc546402177e1e : SUCCESS in 2h 10m 18s\n- openstack-ansible-deploy-aio_metal-ubuntu-bionic https://zuul.opendev.org/t/openstack/build/b17959f745a2478681ec0778a67caa26 : SUCCESS in 1h 35m 01s\n- openstack-ansible-upgrade-aio_metal-centos-7 https://zuul.opendev.org/t/openstack/build/8e73ffedc80b4b01861d3cba13ad2cfc : TIMED_OUT in 3h 06m 06s (non-voting)\n- openstack-ansible-upgrade-aio_metal-ubuntu-bionic https://zuul.opendev.org/t/openstack/build/da53dfa2046b425dacecb49647ec711e : SUCCESS in 2h 05m 33s\n- openstack-ansible-deploy-aio_distro_metal-centos-7 https://zuul.opendev.org/t/openstack/build/0526d92ad58a4ad688761446ea8d4ccf : SUCCESS in 1h 25m 30s\n- openstack-ansible-deploy-aio_distro_metal-ubuntu-bionic https://zuul.opendev.org/t/openstack/build/94659af86fa04d9d80690fa9847e5ae7 : SUCCESS in 54m 34s\n- openstack-ansible-deploy-aio_telemetry_metal-centos-7 https://zuul.opendev.org/t/openstack/build/de8857a212454774bc00e3252ca81dde : SUCCESS in 1h 53m 29s\n- openstack-ansible-deploy-aio_telemetry_metal-debian-buster https://zuul.opendev.org/t/openstack/build/6e200ebc6a0f4fb997cd773f51f3de4d : SUCCESS in 58m 18s\n- openstack-ansible-deploy-aio_telemetry_metal-ubuntu-bionic https://zuul.opendev.org/t/openstack/build/f9b85555fe5143118af3461535561ea9 : SUCCESS in 1h 28m 11s\n- openstack-tox-docs https://zuul.opendev.org/t/openstack/build/7f0e02d09a704d33a741d54e5d6deb3f : SUCCESS in 7m 38s","accounts_in_message":[],"_revision_number":2},{"id":"c07d83e291bf9c5562a9e21cdb778ba1a7830e2f","author":{"_account_id":25023,"name":"Jonathan Rosser","email":"jonathan.rosser@rd.bbc.co.uk","username":"jrosser"},"date":"2020-06-18 06:19:47.000000000","message":"Patch Set 2: Code-Review+2 Workflow+1","accounts_in_message":[],"_revision_number":2},{"id":"24e20ceaa01a3d8028ecb6dcb46af0eeebadafff","tag":"autogenerated:zuul:gate","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2020-06-18 06:19:59.000000000","message":"Patch Set 2: -Verified\n\nStarting gate jobs.","accounts_in_message":[],"_revision_number":2},{"id":"487235ebd9e805a1f9c6f7e92546f13575c3d10d","tag":"autogenerated:zuul:gate","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2020-06-18 08:34:02.000000000","message":"Patch Set 2: Verified+2\n\nBuild succeeded (gate pipeline).\n\n- openstack-ansible-linters https://zuul.opendev.org/t/openstack/build/d781b8e336e74443afafc30f9ff2b257 : SUCCESS in 10m 22s\n- openstack-ansible-deploy-aio_metal-debian-buster https://zuul.opendev.org/t/openstack/build/1719fe9a257546149305f4e11eca7d82 : SUCCESS in 1h 23m 32s\n- openstack-ansible-deploy-aio_metal-centos-7 https://zuul.opendev.org/t/openstack/build/602a17f567e14aa9a1b17e83162a4132 : SUCCESS in 2h 01m 17s\n- openstack-ansible-deploy-aio_metal-ubuntu-bionic https://zuul.opendev.org/t/openstack/build/392b33ceb2624ecbbca6409d3539f789 : SUCCESS in 1h 30m 52s\n- openstack-ansible-deploy-aio_distro_metal-centos-7 https://zuul.opendev.org/t/openstack/build/7775be0396e041d8b90a25d814f70bcd : SUCCESS in 1h 29m 38s\n- openstack-ansible-deploy-aio_distro_metal-ubuntu-bionic https://zuul.opendev.org/t/openstack/build/f5cf6ba430fe4c0cb803a41ccae37c36 : SUCCESS in 1h 10m 03s\n- openstack-ansible-deploy-aio_telemetry_metal-centos-7 https://zuul.opendev.org/t/openstack/build/77b96b6aa83546f1b5efeb4ff7ac302f : SUCCESS in 1h 16m 44s\n- openstack-ansible-deploy-aio_telemetry_metal-debian-buster https://zuul.opendev.org/t/openstack/build/c448afa538804d18b48ee27574d91894 : SUCCESS in 1h 16m 46s\n- openstack-ansible-deploy-aio_telemetry_metal-ubuntu-bionic https://zuul.opendev.org/t/openstack/build/0c2e9fe277474939a99cb14bcd95c444 : SUCCESS in 1h 33m 39s\n- openstack-tox-docs https://zuul.opendev.org/t/openstack/build/c8dbbe07b32149c3b829751fd8ab163f : SUCCESS in 8m 45s","accounts_in_message":[],"_revision_number":2},{"id":"4cfec1b88db924025c0e4df27ee86e4250d9a156","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2020-06-18 08:34:03.000000000","message":"Change has been successfully merged by Zuul","accounts_in_message":[],"_revision_number":2},{"id":"fee6068137ed483407d511ef44381a3fa97f8f88","tag":"autogenerated:zuul:promote","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2020-06-18 08:36:56.000000000","message":"Patch Set 2:\n\nBuild failed (promote pipeline).  For information on how to proceed, see\nhttps://docs.opendev.org/opendev/infra-manual/latest/developers.html#automated-testing\n\n\n- promote-openstack-tox-docs https://zuul.opendev.org/t/openstack/build/f9ba005173304364b0996b499ecda69e : FAILURE in 2m 29s","accounts_in_message":[],"_revision_number":2}],"current_revision_number":2,"current_revision":"bc45afa65065bf9fed2d897fcde7d097224443e0","revisions":{"aae473ecc5a07891e02076abcc30efe4613d7b4d":{"kind":"REWORK","_number":1,"created":"2020-06-17 13:10:51.000000000","uploader":{"_account_id":25023,"name":"Jonathan Rosser","email":"jonathan.rosser@rd.bbc.co.uk","username":"jrosser"},"ref":"refs/changes/06/736206/1","fetch":{"anonymous http":{"url":"https://review.opendev.org/openstack/openstack-ansible-os_aodh","ref":"refs/changes/06/736206/1","commands":{"Checkout":"git fetch https://review.opendev.org/openstack/openstack-ansible-os_aodh refs/changes/06/736206/1 \u0026\u0026 git checkout FETCH_HEAD","Cherry Pick":"git fetch https://review.opendev.org/openstack/openstack-ansible-os_aodh refs/changes/06/736206/1 \u0026\u0026 git cherry-pick FETCH_HEAD","Format Patch":"git fetch https://review.opendev.org/openstack/openstack-ansible-os_aodh refs/changes/06/736206/1 \u0026\u0026 git format-patch -1 --stdout FETCH_HEAD","Pull":"git pull https://review.opendev.org/openstack/openstack-ansible-os_aodh refs/changes/06/736206/1"}}},"commit":{"parents":[{"commit":"e7f2d33644267e9c3abdc75ddd8c8cc6aa45fd1e","subject":"Update TOX_CONSTRAINTS_FILE for stable/ussuri","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/openstack-ansible-os_aodh/commit/e7f2d33644267e9c3abdc75ddd8c8cc6aa45fd1e"}]}],"author":{"name":"Marc Gariepy","email":"gariepy.marc@gmail.com","date":"2020-06-11 18:01:20.000000000","tz":-240},"committer":{"name":"Jonathan Rosser","email":"jonathan.rosser@rd.bbc.co.uk","date":"2020-06-17 13:10:51.000000000","tz":0},"subject":"remove ansible_python_interpreter","message":"remove ansible_python_interpreter\n\nansible_python_interpreter is now set to auto in the tests repo.\nit doesn\u0027t need to be overwritten in the test anymore.\n\nDepends-On: https://review.opendev.org/734676/\nChange-Id: I98386dde8e10e0ded820326af443aa6e70f53321\n(cherry picked from commit e5d245c17de3b4309df5447efac2d3e5a1967f79)\n","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/openstack-ansible-os_aodh/commit/aae473ecc5a07891e02076abcc30efe4613d7b4d"}],"resolve_conflicts_web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/openstack-ansible-os_aodh/commit/aae473ecc5a07891e02076abcc30efe4613d7b4d"}]},"branch":"refs/heads/stable/ussuri"},"bc45afa65065bf9fed2d897fcde7d097224443e0":{"kind":"NO_CODE_CHANGE","_number":2,"created":"2020-06-17 20:17:01.000000000","uploader":{"_account_id":13095,"name":"Marc Gariépy","email":"gariepy.marc@gmail.com","username":"mgariepy"},"ref":"refs/changes/06/736206/2","fetch":{"anonymous http":{"url":"https://review.opendev.org/openstack/openstack-ansible-os_aodh","ref":"refs/changes/06/736206/2","commands":{"Checkout":"git fetch https://review.opendev.org/openstack/openstack-ansible-os_aodh refs/changes/06/736206/2 \u0026\u0026 git checkout FETCH_HEAD","Cherry Pick":"git fetch https://review.opendev.org/openstack/openstack-ansible-os_aodh refs/changes/06/736206/2 \u0026\u0026 git cherry-pick FETCH_HEAD","Format Patch":"git fetch https://review.opendev.org/openstack/openstack-ansible-os_aodh refs/changes/06/736206/2 \u0026\u0026 git format-patch -1 --stdout FETCH_HEAD","Pull":"git pull https://review.opendev.org/openstack/openstack-ansible-os_aodh refs/changes/06/736206/2"}}},"commit":{"parents":[{"commit":"e7f2d33644267e9c3abdc75ddd8c8cc6aa45fd1e","subject":"Update TOX_CONSTRAINTS_FILE for stable/ussuri","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/openstack-ansible-os_aodh/commit/e7f2d33644267e9c3abdc75ddd8c8cc6aa45fd1e"}]}],"author":{"name":"Marc Gariepy","email":"gariepy.marc@gmail.com","date":"2020-06-11 18:01:20.000000000","tz":-240},"committer":{"name":"Marc Gariépy (mgariepy)","email":"gariepy.marc@gmail.com","date":"2020-06-17 20:16:57.000000000","tz":0},"subject":"remove ansible_python_interpreter","message":"remove ansible_python_interpreter\n\nansible_python_interpreter is now set to auto in the tests repo.\nit doesn\u0027t need to be overwritten in the test anymore.\n\nChange-Id: I98386dde8e10e0ded820326af443aa6e70f53321\n(cherry picked from commit e5d245c17de3b4309df5447efac2d3e5a1967f79)\n","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/openstack-ansible-os_aodh/commit/bc45afa65065bf9fed2d897fcde7d097224443e0"}],"resolve_conflicts_web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/openstack-ansible-os_aodh/commit/bc45afa65065bf9fed2d897fcde7d097224443e0"}]},"branch":"refs/heads/stable/ussuri"}},"requirements":[],"submit_records":[],"submit_requirements":[]}
