)]}'
{"id":"openstack%2Fopenstack-ansible-os_aodh~739401","triplet_id":"openstack%2Fopenstack-ansible-os_aodh~master~Ie16ad8e0cac74fe2d92663bc9ddb8d04c21c2878","project":"openstack/openstack-ansible-os_aodh","branch":"master","topic":"osa/el8","hashtags":[],"change_id":"Ie16ad8e0cac74fe2d92663bc9ddb8d04c21c2878","subject":"Add centos-8 support","status":"MERGED","created":"2020-07-06 05:23:43.000000000","updated":"2020-07-27 18:37:54.000000000","submitted":"2020-07-25 17:09:10.000000000","submitter":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"total_comment_count":0,"unresolved_comment_count":0,"has_review_started":true,"submission_id":"739401-1595696950630-38a1cb2b","meta_rev_id":"03cfe41fa637111f950c672ea8075387431efc03","_number":739401,"virtual_id_number":739401,"owner":{"_account_id":28752,"name":"Satish Patel","email":"satish.txt@gmail.com","username":"spatel"},"actions":{},"labels":{"Verified":{"approved":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"all":[{"value":0,"date":"2020-07-23 03:22:47.000000000","_account_id":28752,"name":"Satish Patel","email":"satish.txt@gmail.com","username":"spatel"},{"value":0,"_account_id":28619,"name":"Dmitriy Rabotyagov","email":"noonedeadpunk@gmail.com","username":"noonedeadpunk"},{"tag":"autogenerated:zuul:gate","value":2,"date":"2020-07-25 17:09:10.000000000","post_submit":true,"permitted_voting_range":{"min":2,"max":2},"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},{"value":0,"_account_id":25023,"name":"Jonathan Rosser","email":"jonathan.rosser@rd.bbc.co.uk","username":"jrosser"}],"values":{"-2":"Fails","-1":"Doesn\u0027t seem to work"," 0":"No score","+1":"Works for me","+2":"Verified"},"description":"","default_value":0,"optional":true},"Code-Review":{"approved":{"_account_id":28619,"name":"Dmitriy Rabotyagov","email":"noonedeadpunk@gmail.com","username":"noonedeadpunk"},"all":[{"value":0,"_account_id":28752,"name":"Satish Patel","email":"satish.txt@gmail.com","username":"spatel"},{"value":2,"date":"2020-07-24 14:21:21.000000000","permitted_voting_range":{"min":2,"max":2},"_account_id":28619,"name":"Dmitriy Rabotyagov","email":"noonedeadpunk@gmail.com","username":"noonedeadpunk"},{"value":0,"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},{"value":2,"date":"2020-07-24 14:24:01.000000000","permitted_voting_range":{"min":2,"max":2},"_account_id":25023,"name":"Jonathan Rosser","email":"jonathan.rosser@rd.bbc.co.uk","username":"jrosser"}],"values":{"-2":"Do not merge","-1":"This patch needs further work before it can be merged"," 0":"No score","+1":"Looks good to me, but someone else must approve","+2":"Looks good to me (core reviewer)"},"description":"","default_value":0,"optional":true},"Workflow":{"approved":{"_account_id":25023,"name":"Jonathan Rosser","email":"jonathan.rosser@rd.bbc.co.uk","username":"jrosser"},"all":[{"value":0,"_account_id":28752,"name":"Satish Patel","email":"satish.txt@gmail.com","username":"spatel"},{"value":0,"_account_id":28619,"name":"Dmitriy Rabotyagov","email":"noonedeadpunk@gmail.com","username":"noonedeadpunk"},{"value":0,"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},{"value":1,"date":"2020-07-24 14:24:01.000000000","permitted_voting_range":{"min":1,"max":1},"_account_id":25023,"name":"Jonathan Rosser","email":"jonathan.rosser@rd.bbc.co.uk","username":"jrosser"}],"values":{"-1":"Work in progress"," 0":"Ready for reviews","+1":"Approved"},"description":"","default_value":0,"optional":true},"Backport-Candidate":{"all":[{"value":0,"_account_id":28752,"name":"Satish Patel","email":"satish.txt@gmail.com","username":"spatel"},{"value":0,"_account_id":28619,"name":"Dmitriy Rabotyagov","email":"noonedeadpunk@gmail.com","username":"noonedeadpunk"},{"value":0,"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},{"value":0,"_account_id":25023,"name":"Jonathan Rosser","email":"jonathan.rosser@rd.bbc.co.uk","username":"jrosser"}],"values":{"-1":"Not A Backport Candidate"," 0":"Backport Review Needed","+1":"Proposed Backport"},"description":"","default_value":0,"optional":true}},"removable_reviewers":[],"reviewers":{"REVIEWER":[{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},{"_account_id":25023,"name":"Jonathan Rosser","email":"jonathan.rosser@rd.bbc.co.uk","username":"jrosser"},{"_account_id":28619,"name":"Dmitriy Rabotyagov","email":"noonedeadpunk@gmail.com","username":"noonedeadpunk"},{"_account_id":28752,"name":"Satish Patel","email":"satish.txt@gmail.com","username":"spatel"}]},"pending_reviewers":{},"reviewer_updates":[{"updated":"2020-07-24 14:21:21.000000000","updated_by":{"_account_id":28619,"name":"Dmitriy Rabotyagov","email":"noonedeadpunk@gmail.com","username":"noonedeadpunk"},"reviewer":{"_account_id":28619,"name":"Dmitriy Rabotyagov","email":"noonedeadpunk@gmail.com","username":"noonedeadpunk"},"state":"REVIEWER"},{"updated":"2020-07-24 14:24:01.000000000","updated_by":{"_account_id":25023,"name":"Jonathan Rosser","email":"jonathan.rosser@rd.bbc.co.uk","username":"jrosser"},"reviewer":{"_account_id":25023,"name":"Jonathan Rosser","email":"jonathan.rosser@rd.bbc.co.uk","username":"jrosser"},"state":"REVIEWER"},{"updated":"2020-07-25 17:09:10.000000000","updated_by":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"reviewer":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"state":"REVIEWER"}],"messages":[{"id":"b4f21eca4fa6457148f33ec988a42a01d4449e66","author":{"_account_id":28752,"name":"Satish Patel","email":"satish.txt@gmail.com","username":"spatel"},"date":"2020-07-06 05:23:43.000000000","message":"Uploaded patch set 1.","accounts_in_message":[],"_revision_number":1},{"id":"27d89314a3d807deb0d444a1383a5367b9c3912a","tag":"autogenerated:zuul:check","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2020-07-06 08:30:14.000000000","message":"Patch Set 1: Verified-1\n\nBuild failed (check pipeline).  For information on how to proceed, see\nhttps://docs.opendev.org/opendev/infra-manual/latest/developers.html#automated-testing\n\n\n- openstack-ansible-linters https://zuul.opendev.org/t/openstack/build/70732ac2a75f49f4909481ddc60ff193 : SUCCESS in 7m 51s\n- openstack-ansible-deploy-aio_metal-debian-buster https://zuul.opendev.org/t/openstack/build/e08b465a4da94b4a867a89a35da6647b : SUCCESS in 58m 45s\n- openstack-ansible-deploy-aio_metal-centos-7 https://zuul.opendev.org/t/openstack/build/bf5a141d62614b259e7aa80c6f6e5b8a : SUCCESS in 1h 52m 47s\n- openstack-ansible-deploy-aio_metal-ubuntu-bionic https://zuul.opendev.org/t/openstack/build/666670fcc77246e68838ed7c4080b6d7 : SUCCESS in 1h 29m 16s\n- openstack-ansible-deploy-aio_metal-ubuntu-focal https://zuul.opendev.org/t/openstack/build/783dfcb51d904f3091381995a22f7942 : FAILURE in 58m 05s\n- openstack-ansible-upgrade-aio_metal-centos-7 https://zuul.opendev.org/t/openstack/build/df340026544444d1801453907e256c53 : TIMED_OUT in 3h 04m 38s (non-voting)\n- openstack-ansible-upgrade-aio_metal-ubuntu-bionic https://zuul.opendev.org/t/openstack/build/65cb6254af4e4ad9909f8d15aade0146 : SUCCESS in 2h 19m 27s\n- openstack-ansible-deploy-aio_distro_metal-centos-7 https://zuul.opendev.org/t/openstack/build/1696770aa434497c9b820d3fe8d1d967 : SUCCESS in 1h 22m 28s\n- openstack-ansible-deploy-aio_distro_metal-ubuntu-bionic https://zuul.opendev.org/t/openstack/build/4c474d6b590644d1acd8c399cab3b55d : SUCCESS in 1h 17m 10s\n- openstack-ansible-deploy-aio_telemetry_metal-centos-7 https://zuul.opendev.org/t/openstack/build/010554f9a33c4882976ca28ba0dfd847 : SUCCESS in 1h 05m 13s\n- openstack-ansible-deploy-aio_telemetry_metal-debian-buster https://zuul.opendev.org/t/openstack/build/95211c17b0bd45ea94acdf4b2044447e : SUCCESS in 1h 19m 54s\n- openstack-ansible-deploy-aio_telemetry_metal-ubuntu-bionic https://zuul.opendev.org/t/openstack/build/bf101f8c18f54832a7b9df6d21695fbf : SUCCESS in 1h 32m 16s\n- openstack-tox-docs https://zuul.opendev.org/t/openstack/build/f833df646dfa45b5861d8f4bfdc88e90 : SUCCESS in 5m 12s","accounts_in_message":[],"_revision_number":1},{"id":"dc874f31bbedffc1edebdd5ac593799696c1e8ee","author":{"_account_id":28752,"name":"Satish Patel","email":"satish.txt@gmail.com","username":"spatel"},"date":"2020-07-07 04:54:57.000000000","message":"Uploaded patch set 2.","accounts_in_message":[],"_revision_number":2},{"id":"45f26deaa55cd879d7eb37a84b0e4c992744d02a","tag":"autogenerated:zuul:check","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2020-07-07 08:03:16.000000000","message":"Patch Set 2: Verified-1\n\nBuild failed (check pipeline).  For information on how to proceed, see\nhttps://docs.opendev.org/opendev/infra-manual/latest/developers.html#automated-testing\n\n\n- openstack-ansible-linters https://zuul.opendev.org/t/openstack/build/1d6af82440054195a8806a175fdec85d : SUCCESS in 7m 36s\n- openstack-ansible-deploy-aio_metal-debian-buster https://zuul.opendev.org/t/openstack/build/8319c62187f346edabaf55cfd3226cfd : SUCCESS in 1h 19m 52s\n- openstack-ansible-deploy-aio_metal-centos-7 https://zuul.opendev.org/t/openstack/build/c7c7268f70d54536970cb4c40b371bca : SUCCESS in 1h 45m 40s\n- openstack-ansible-deploy-aio_metal-ubuntu-bionic https://zuul.opendev.org/t/openstack/build/8ad0075713a0432a84b2d3c9c5a1c999 : SUCCESS in 1h 26m 16s\n- openstack-ansible-deploy-aio_metal-ubuntu-focal https://zuul.opendev.org/t/openstack/build/424b062ac9154bba8feaf7949785ba6b : FAILURE in 55m 19s\n- openstack-ansible-upgrade-aio_metal-centos-7 https://zuul.opendev.org/t/openstack/build/eaa6f42be6104f2dabf4148321a0de5d : SUCCESS in 2h 59m 28s (non-voting)\n- openstack-ansible-upgrade-aio_metal-ubuntu-bionic https://zuul.opendev.org/t/openstack/build/fdaf434096c0409cb21df8fed8363898 : SUCCESS in 2h 20m 13s\n- openstack-ansible-deploy-aio_distro_metal-centos-7 https://zuul.opendev.org/t/openstack/build/7a1633030be947858e9100d61c3f5d82 : SUCCESS in 1h 15m 52s\n- openstack-ansible-deploy-aio_distro_metal-ubuntu-bionic https://zuul.opendev.org/t/openstack/build/69c5a4f2ad01495cbb20c84a02cc4dab : SUCCESS in 1h 08m 48s\n- openstack-ansible-deploy-aio_telemetry_metal-centos-7 https://zuul.opendev.org/t/openstack/build/ee9a37d398bc40f8827b320da6de71b8 : SUCCESS in 1h 43m 59s\n- openstack-ansible-deploy-aio_telemetry_metal-debian-buster https://zuul.opendev.org/t/openstack/build/ebe8ea0d154c40558371fc3260cd4b15 : TIMED_OUT in 3h 06m 15s\n- openstack-ansible-deploy-aio_telemetry_metal-ubuntu-bionic https://zuul.opendev.org/t/openstack/build/80f6493701c14beea4a3244c3d3e3b66 : SUCCESS in 2h 50m 14s\n- openstack-tox-docs https://zuul.opendev.org/t/openstack/build/7251c3e205e04f5d92b27f40ae969b4d : SUCCESS in 7m 20s","accounts_in_message":[],"_revision_number":2},{"id":"42b4d5c5e18f49e3589ea5aa469e2851b2ff4604","author":{"_account_id":28752,"name":"Satish Patel","email":"satish.txt@gmail.com","username":"spatel"},"date":"2020-07-09 13:45:29.000000000","message":"Patch Set 2:\n\nrecheck","accounts_in_message":[],"_revision_number":2},{"id":"7e1b1af25959f8394010ad99854d6e4afb70c5cb","tag":"autogenerated:zuul:check","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2020-07-09 16:59:06.000000000","message":"Patch Set 2:\n\nBuild failed (check pipeline).  For information on how to proceed, see\nhttps://docs.opendev.org/opendev/infra-manual/latest/developers.html#automated-testing\n\n\n- openstack-ansible-linters https://zuul.opendev.org/t/openstack/build/e89da140cad1477b88ff1259546c34f8 : SUCCESS in 9m 06s\n- openstack-ansible-deploy-aio_metal-debian-buster https://zuul.opendev.org/t/openstack/build/01c1a093055c41abb2c53424a4b00e37 : SUCCESS in 1h 17m 14s\n- openstack-ansible-deploy-aio_metal-centos-7 https://zuul.opendev.org/t/openstack/build/b83d06c1204a452e9d1b145474a64c49 : SUCCESS in 1h 41m 30s\n- openstack-ansible-deploy-aio_metal-centos-8 https://zuul.opendev.org/t/openstack/build/d40de3d7b83242cdb8f8a02a2bf1c91a : FAILURE in 49m 26s\n- openstack-ansible-deploy-aio_metal-ubuntu-bionic https://zuul.opendev.org/t/openstack/build/71eb64e59b0f45ffaa11f0a3622bb5d5 : SUCCESS in 1h 10m 50s\n- openstack-ansible-deploy-aio_metal-ubuntu-focal https://zuul.opendev.org/t/openstack/build/15afdd2f4c7e4cb498013c6eeddced7c : FAILURE in 1h 01m 34s\n- openstack-ansible-upgrade-aio_metal-centos-7 https://zuul.opendev.org/t/openstack/build/5c4c4fc64e93471883f7f0fa99a3de7b : TIMED_OUT in 3h 04m 01s (non-voting)\n- openstack-ansible-upgrade-aio_metal-ubuntu-bionic https://zuul.opendev.org/t/openstack/build/79a0914be416423f892c67ec49a69cdb : SUCCESS in 2h 29m 01s\n- openstack-ansible-deploy-aio_distro_metal-centos-7 https://zuul.opendev.org/t/openstack/build/a123188ec39c44a5921b313363ca4700 : SUCCESS in 1h 22m 54s\n- openstack-ansible-deploy-aio_distro_metal-centos-8 https://zuul.opendev.org/t/openstack/build/432c2e822c784b60a259084fd4495d61 : FAILURE in 1h 03m 08s\n- openstack-ansible-deploy-aio_distro_metal-ubuntu-bionic https://zuul.opendev.org/t/openstack/build/98661cb617ec4731af8de4c32323b54d : SUCCESS in 1h 33m 02s\n- openstack-ansible-deploy-aio_telemetry_metal-centos-7 https://zuul.opendev.org/t/openstack/build/410546d174d34d90b3ff513c363cdae0 : SUCCESS in 1h 44m 41s\n- openstack-ansible-deploy-aio_telemetry_metal-debian-buster https://zuul.opendev.org/t/openstack/build/a05254ad415b4279a1a3fd6b4fed0413 : SUCCESS in 1h 18m 54s\n- openstack-ansible-deploy-aio_telemetry_metal-ubuntu-bionic https://zuul.opendev.org/t/openstack/build/53a51836415a40c481d66a0291e3032d : SUCCESS in 1h 30m 54s\n- openstack-tox-docs https://zuul.opendev.org/t/openstack/build/5e934f2e9e5046ea9b3da62d9aad92e4 : SUCCESS in 7m 51s","accounts_in_message":[],"_revision_number":2},{"id":"d6807ac57c2024c18ef68a5ca892a3a78815fb64","author":{"_account_id":28752,"name":"Satish Patel","email":"satish.txt@gmail.com","username":"spatel"},"date":"2020-07-09 19:05:11.000000000","message":"Patch Set 2:\n\nrecheck","accounts_in_message":[],"_revision_number":2},{"id":"bd4fd797db846b255da70ea0b7fac13504066107","tag":"autogenerated:zuul:check","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2020-07-09 21:43:56.000000000","message":"Patch Set 2:\n\nBuild failed (check pipeline).  For information on how to proceed, see\nhttps://docs.opendev.org/opendev/infra-manual/latest/developers.html#automated-testing\n\n\n- openstack-ansible-linters https://zuul.opendev.org/t/openstack/build/4b1f8f703bfe419faede43ff966d96cb : SUCCESS in 8m 54s\n- openstack-ansible-deploy-aio_metal-debian-buster https://zuul.opendev.org/t/openstack/build/9345aa7a03cd4eab82f91a031e0ea8ee : SUCCESS in 1h 23m 03s\n- openstack-ansible-deploy-aio_metal-centos-7 https://zuul.opendev.org/t/openstack/build/13d73d2fd6df40218df6dc149fb882af : SUCCESS in 1h 48m 25s\n- openstack-ansible-deploy-aio_metal-centos-8 https://zuul.opendev.org/t/openstack/build/0d8a1bffb4664177b06df48962474f18 : FAILURE in 1h 11m 39s\n- openstack-ansible-deploy-aio_metal-ubuntu-bionic https://zuul.opendev.org/t/openstack/build/a189ae9de11a41a0a8f90ad4f472ca25 : SUCCESS in 1h 26m 04s\n- openstack-ansible-deploy-aio_metal-ubuntu-focal https://zuul.opendev.org/t/openstack/build/55cba466d3074c039e36854a6af4457b : FAILURE in 43m 19s\n- openstack-ansible-upgrade-aio_metal-centos-7 https://zuul.opendev.org/t/openstack/build/65b19daaa9d643669e569cac6d29a2bc : SUCCESS in 2h 33m 50s (non-voting)\n- openstack-ansible-upgrade-aio_metal-ubuntu-bionic https://zuul.opendev.org/t/openstack/build/3d32eeb6b3304436a7621c279b7d0a30 : SUCCESS in 2h 13m 34s\n- openstack-ansible-deploy-aio_distro_metal-centos-7 https://zuul.opendev.org/t/openstack/build/fae57031e717429ab5c79723cdf74753 : SUCCESS in 1h 32m 08s\n- openstack-ansible-deploy-aio_distro_metal-centos-8 https://zuul.opendev.org/t/openstack/build/b41ab9ef5d414cbe87261afea3742e05 : FAILURE in 49m 09s\n- openstack-ansible-deploy-aio_distro_metal-ubuntu-bionic https://zuul.opendev.org/t/openstack/build/27970e8e6b924fba873c121aec6b5086 : SUCCESS in 1h 04m 31s\n- openstack-ansible-deploy-aio_telemetry_metal-centos-7 https://zuul.opendev.org/t/openstack/build/c440549feb554deb9fa41d5594fdd8c8 : SUCCESS in 1h 44m 55s\n- openstack-ansible-deploy-aio_telemetry_metal-debian-buster https://zuul.opendev.org/t/openstack/build/7bd50c93ffb54c91a998bbc1fb7bc4a6 : SUCCESS in 1h 16m 16s\n- openstack-ansible-deploy-aio_telemetry_metal-ubuntu-bionic https://zuul.opendev.org/t/openstack/build/722e207caaf5445f91f1f8a3a1acdb72 : SUCCESS in 1h 23m 37s\n- openstack-tox-docs https://zuul.opendev.org/t/openstack/build/acd6c47bc0cf496a8e1e7124c59662b5 : SUCCESS in 7m 01s","accounts_in_message":[],"_revision_number":2},{"id":"150f4f5187ed18b5a5aa33f3efa419381914e941","author":{"_account_id":28752,"name":"Satish Patel","email":"satish.txt@gmail.com","username":"spatel"},"date":"2020-07-10 15:52:47.000000000","message":"Patch Set 2:\n\nrecheck","accounts_in_message":[],"_revision_number":2},{"id":"438f6fbbc9b8ca4ce8b247ed01c5a84c9b21f39b","tag":"autogenerated:zuul:check","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2020-07-10 18:49:18.000000000","message":"Patch Set 2:\n\nBuild failed (check pipeline).  For information on how to proceed, see\nhttps://docs.opendev.org/opendev/infra-manual/latest/developers.html#automated-testing\n\n\n- openstack-ansible-linters https://zuul.opendev.org/t/openstack/build/fd9e1936594045e1a8575beb59bfeaf5 : SUCCESS in 9m 38s\n- openstack-ansible-deploy-aio_metal-debian-buster https://zuul.opendev.org/t/openstack/build/bbf590f96e804f40811e08d04654d1a3 : SUCCESS in 1h 16m 56s\n- openstack-ansible-deploy-aio_metal-centos-7 https://zuul.opendev.org/t/openstack/build/172ff3931efd4d88a434fb6489a7b70e : SUCCESS in 1h 50m 55s\n- openstack-ansible-deploy-aio_metal-centos-8 https://zuul.opendev.org/t/openstack/build/cfd07339127f4b77925a4926522237f3 : FAILURE in 1h 11m 18s\n- openstack-ansible-deploy-aio_metal-ubuntu-bionic https://zuul.opendev.org/t/openstack/build/4af57b9d90934671a2e4af26c7ff114c : SUCCESS in 1h 18m 57s\n- openstack-ansible-deploy-aio_metal-ubuntu-focal https://zuul.opendev.org/t/openstack/build/b98eff75da974ffc97d7b591da19a366 : FAILURE in 1h 01m 53s\n- openstack-ansible-upgrade-aio_metal-centos-7 https://zuul.opendev.org/t/openstack/build/d09e1d9f0c024004ab92ac0a7d2c1c70 : SUCCESS in 2h 49m 12s (non-voting)\n- openstack-ansible-upgrade-aio_metal-ubuntu-bionic https://zuul.opendev.org/t/openstack/build/a05d5550f2284f4bbb758aced6f4bce0 : SUCCESS in 2h 15m 18s\n- openstack-ansible-deploy-aio_distro_metal-centos-7 https://zuul.opendev.org/t/openstack/build/06417d313aa64fd3b82eecfb7ddbf99c : SUCCESS in 1h 22m 10s\n- openstack-ansible-deploy-aio_distro_metal-centos-8 https://zuul.opendev.org/t/openstack/build/dfcf49ead4de49e0ba02e9627f44d403 : FAILURE in 47m 08s\n- openstack-ansible-deploy-aio_distro_metal-ubuntu-bionic https://zuul.opendev.org/t/openstack/build/2f3527cc61d8424f856128cdf84beb78 : SUCCESS in 1h 46m 35s\n- openstack-ansible-deploy-aio_telemetry_metal-centos-7 https://zuul.opendev.org/t/openstack/build/80f9b03022a04b3a812497e3657d4c8c : SUCCESS in 1h 46m 07s\n- openstack-ansible-deploy-aio_telemetry_metal-debian-buster https://zuul.opendev.org/t/openstack/build/132ea94170f644749efc3414e92302d3 : SUCCESS in 1h 07m 57s\n- openstack-ansible-deploy-aio_telemetry_metal-ubuntu-bionic https://zuul.opendev.org/t/openstack/build/bda03797245e4971aa0924c560100e46 : SUCCESS in 1h 49m 00s\n- openstack-tox-docs https://zuul.opendev.org/t/openstack/build/5a3f9de1558a4993bf9c1ed0b7f452ba : SUCCESS in 7m 29s","accounts_in_message":[],"_revision_number":2},{"id":"7d5e19201339674b8d2ad8f70b0746e05ad3998f","author":{"_account_id":28752,"name":"Satish Patel","email":"satish.txt@gmail.com","username":"spatel"},"date":"2020-07-20 00:32:35.000000000","message":"Uploaded patch set 3: Commit message was updated.","accounts_in_message":[],"_revision_number":3},{"id":"2bef44409916be309d2dee67b6b38928cbc91a1d","tag":"autogenerated:zuul:check","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2020-07-20 00:34:05.000000000","message":"Patch Set 3: Verified-1\n\nThis change depends on a change that failed to merge.","accounts_in_message":[],"_revision_number":3},{"id":"07d70d0e917d66ee3a44a979e258985204497edc","author":{"_account_id":28752,"name":"Satish Patel","email":"satish.txt@gmail.com","username":"spatel"},"date":"2020-07-22 17:16:40.000000000","message":"Uploaded patch set 4.","accounts_in_message":[],"_revision_number":4},{"id":"d0333bee568ccb3148a8ad8d2bbd050e12a4e96a","tag":"autogenerated:zuul:check","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2020-07-22 20:49:06.000000000","message":"Patch Set 4: Verified-1\n\nBuild failed (check pipeline).  For information on how to proceed, see\nhttps://docs.opendev.org/opendev/infra-manual/latest/developers.html#automated-testing\n\n\n- openstack-ansible-linters https://zuul.opendev.org/t/openstack/build/8e83d300cb0d4a14b342f1df084def99 : SUCCESS in 8m 51s\n- openstack-ansible-deploy-aio_metal-debian-buster https://zuul.opendev.org/t/openstack/build/a3080ce5405f440db5fe1345c4ab28d2 : SUCCESS in 1h 04m 44s\n- openstack-ansible-deploy-aio_metal-centos-7 https://zuul.opendev.org/t/openstack/build/131c38ed3eba47fca97275254731fa83 : SUCCESS in 1h 14m 53s\n- openstack-ansible-deploy-aio_metal-centos-8 https://zuul.opendev.org/t/openstack/build/4c5dd9d568134c998d74e959724cdfdb : FAILURE in 1h 07m 08s (non-voting)\n- openstack-ansible-deploy-aio_metal-ubuntu-bionic https://zuul.opendev.org/t/openstack/build/cc09af4032d84de8abb0abc0269a96d8 : SUCCESS in 1h 11m 53s\n- openstack-ansible-deploy-aio_metal-ubuntu-focal https://zuul.opendev.org/t/openstack/build/9dcec4f8de8e42ddbbf73b4ca70c5a01 : FAILURE in 1h 00m 42s (non-voting)\n- openstack-ansible-upgrade-aio_metal-centos-7 https://zuul.opendev.org/t/openstack/build/99ce880277af4fa2a9a34856fd5e88f8 : TIMED_OUT in 3h 07m 44s (non-voting)\n- openstack-ansible-upgrade-aio_metal-ubuntu-bionic https://zuul.opendev.org/t/openstack/build/0c82a54c481b4819bf75d6c98f1619f0 : SUCCESS in 2h 24m 47s\n- openstack-ansible-deploy-aio_distro_metal-centos-7 https://zuul.opendev.org/t/openstack/build/059f5793fdf54158a5a5a136dd1b6cdf : TIMED_OUT in 3h 12m 23s\n- openstack-ansible-deploy-aio_distro_metal-centos-8 https://zuul.opendev.org/t/openstack/build/d45ae02902424d349cba86c12f2a6be6 : FAILURE in 45m 04s (non-voting)\n- openstack-ansible-deploy-aio_distro_metal-ubuntu-bionic https://zuul.opendev.org/t/openstack/build/c05c44aeb75f48beb05b11c97fa2190d : SUCCESS in 1h 07m 43s\n- openstack-ansible-deploy-aio_telemetry_metal-centos-7 https://zuul.opendev.org/t/openstack/build/2981aca133814d129ccfb3f0e0755854 : SUCCESS in 1h 36m 21s\n- openstack-ansible-deploy-aio_telemetry_metal-debian-buster https://zuul.opendev.org/t/openstack/build/b8123def97a84fdca51baa82391ec9e4 : SUCCESS in 1h 22m 58s\n- openstack-ansible-deploy-aio_telemetry_metal-ubuntu-bionic https://zuul.opendev.org/t/openstack/build/0cc8d36384894267a4c3e5d75eb128ad : SUCCESS in 1h 22m 45s\n- openstack-tox-docs https://zuul.opendev.org/t/openstack/build/3321c4888b874fdda142832c0c7066a8 : SUCCESS in 6m 02s","accounts_in_message":[],"_revision_number":4},{"id":"ebf7ecdbe5bd2b0669ae32681e01a5575ab0c3dc","author":{"_account_id":28752,"name":"Satish Patel","email":"satish.txt@gmail.com","username":"spatel"},"date":"2020-07-23 03:22:47.000000000","message":"Patch Set 4:\n\nrecheck","accounts_in_message":[],"_revision_number":4},{"id":"93d1be8666e7173437ef0f601b9ac385b71ae388","tag":"autogenerated:zuul:check","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2020-07-23 06:31:14.000000000","message":"Patch Set 4: Verified+1\n\nBuild succeeded (check pipeline).\n\n- openstack-ansible-linters https://zuul.opendev.org/t/openstack/build/cfa14dd9bf5f4998b60c9ba25e4015d3 : SUCCESS in 6m 07s\n- openstack-ansible-deploy-aio_metal-debian-buster https://zuul.opendev.org/t/openstack/build/20179c3223464395818d39acb6a23514 : SUCCESS in 1h 10m 37s\n- openstack-ansible-deploy-aio_metal-centos-7 https://zuul.opendev.org/t/openstack/build/1a7ec46a3b47428fa9d97b0afc7b7002 : SUCCESS in 1h 51m 33s\n- openstack-ansible-deploy-aio_metal-centos-8 https://zuul.opendev.org/t/openstack/build/040ef3c4121743eba280d919d905ecc7 : FAILURE in 1h 03m 34s (non-voting)\n- openstack-ansible-deploy-aio_metal-ubuntu-bionic https://zuul.opendev.org/t/openstack/build/09298bb6c53d4316bf74909e3ad4b07a : SUCCESS in 1h 35m 43s\n- openstack-ansible-deploy-aio_metal-ubuntu-focal https://zuul.opendev.org/t/openstack/build/0f67bdb7f6434156881b39d248f709a9 : FAILURE in 52m 41s (non-voting)\n- openstack-ansible-upgrade-aio_metal-centos-7 https://zuul.opendev.org/t/openstack/build/605bf36a41dd4793bfa78ee955973faf : TIMED_OUT in 3h 04m 28s (non-voting)\n- openstack-ansible-upgrade-aio_metal-ubuntu-bionic https://zuul.opendev.org/t/openstack/build/a0247fe802b848dd801cb7a73d95bd82 : SUCCESS in 2h 13m 53s\n- openstack-ansible-deploy-aio_distro_metal-centos-7 https://zuul.opendev.org/t/openstack/build/d88ac17763124880bd2fc124a43934cf : SUCCESS in 1h 09m 41s\n- openstack-ansible-deploy-aio_distro_metal-centos-8 https://zuul.opendev.org/t/openstack/build/5a85828857174e80ad859ef8d22f4bbf : FAILURE in 45m 46s (non-voting)\n- openstack-ansible-deploy-aio_distro_metal-ubuntu-bionic https://zuul.opendev.org/t/openstack/build/7d13d5ad611746caa4076647d3fea0d1 : SUCCESS in 1h 10m 48s\n- openstack-ansible-deploy-aio_telemetry_metal-centos-7 https://zuul.opendev.org/t/openstack/build/5a9d44bc510a4b34ba9ca7f226110e41 : SUCCESS in 1h 44m 13s\n- openstack-ansible-deploy-aio_telemetry_metal-debian-buster https://zuul.opendev.org/t/openstack/build/c6dd1e5991914593bae26fd8995ae1f6 : SUCCESS in 1h 18m 46s\n- openstack-ansible-deploy-aio_telemetry_metal-ubuntu-bionic https://zuul.opendev.org/t/openstack/build/2bab9f0b8b30483ca85b8b41083ea3f1 : SUCCESS in 1h 23m 28s\n- openstack-tox-docs https://zuul.opendev.org/t/openstack/build/2363f6df34e5440eac8f60c7c3dfcc3d : SUCCESS in 4m 38s","accounts_in_message":[],"_revision_number":4},{"id":"0ce07b71bba9e4ee86b64e0699c561bb6bdb4016","author":{"_account_id":28619,"name":"Dmitriy Rabotyagov","email":"noonedeadpunk@gmail.com","username":"noonedeadpunk"},"date":"2020-07-24 14:21:21.000000000","message":"Patch Set 4: Code-Review+2","accounts_in_message":[],"_revision_number":4},{"id":"95e2256eb9f960f82a978a2c61caad5cb21cf7fe","author":{"_account_id":25023,"name":"Jonathan Rosser","email":"jonathan.rosser@rd.bbc.co.uk","username":"jrosser"},"date":"2020-07-24 14:24:01.000000000","message":"Patch Set 4: Code-Review+2 Workflow+1","accounts_in_message":[],"_revision_number":4},{"id":"570e4200205ffbf90e999a0e04472f3ec8bf3fe1","tag":"autogenerated:zuul:gate","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2020-07-24 14:24:23.000000000","message":"Patch Set 4: -Verified\n\nStarting gate jobs.","accounts_in_message":[],"_revision_number":4},{"id":"590b9c40e19efcb4e3f45ca0f9b83369b8e42860","tag":"autogenerated:zuul:gate","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2020-07-24 16:40:10.000000000","message":"Patch Set 4: Verified-2\n\nBuild failed (gate pipeline).  For information on how to proceed, see\nhttps://docs.opendev.org/opendev/infra-manual/latest/developers.html#automated-testing\n\n\n- openstack-ansible-linters https://zuul.opendev.org/t/openstack/build/8fe63f047aa2406e94a2520528f89261 : SUCCESS in 8m 59s\n- openstack-ansible-deploy-aio_metal-debian-buster https://zuul.opendev.org/t/openstack/build/c94071a3987e4f09bc8f9d2e5a2c18e1 : SUCCESS in 55m 14s\n- openstack-ansible-deploy-aio_metal-centos-7 https://zuul.opendev.org/t/openstack/build/4df70f58b10f446b9aae4d14b6ef657d : SUCCESS in 1h 58m 16s\n- openstack-ansible-deploy-aio_lxc-centos-8 https://zuul.opendev.org/t/openstack/build/8c1377ef2e294366b5db925adbcf417d : FAILURE in 2h 01m 32s\n- openstack-ansible-deploy-aio_metal-ubuntu-bionic https://zuul.opendev.org/t/openstack/build/bb723ed6a7ce402b8068822a52c1b657 : SUCCESS in 1h 34m 44s\n- openstack-ansible-deploy-aio_metal-ubuntu-focal https://zuul.opendev.org/t/openstack/build/40f5ae2d14354966bb3456564eba10ff : FAILURE in 1h 12m 06s (non-voting)\n- openstack-ansible-deploy-aio_distro_metal-centos-7 https://zuul.opendev.org/t/openstack/build/0a7e1a95b70e47fd9d2896f62e04d6ba : SUCCESS in 1h 23m 11s\n- openstack-ansible-deploy-aio_distro_metal-centos-8 https://zuul.opendev.org/t/openstack/build/2c05878394624e8794d581ace42b7d64 : FAILURE in 50m 43s (non-voting)\n- openstack-ansible-deploy-aio_distro_metal-ubuntu-bionic https://zuul.opendev.org/t/openstack/build/d1ffcce7ef204107ac4ef4a789a24532 : SUCCESS in 1h 33m 01s\n- openstack-ansible-deploy-aio_telemetry_metal-centos-7 https://zuul.opendev.org/t/openstack/build/c67307c83cce40baa48e03c959849158 : SUCCESS in 1h 34m 05s\n- openstack-ansible-deploy-aio_telemetry_metal-debian-buster https://zuul.opendev.org/t/openstack/build/0a2b2cedca0543768aab5efc14e5167e : SUCCESS in 1h 18m 14s\n- openstack-ansible-deploy-aio_telemetry_metal-ubuntu-bionic https://zuul.opendev.org/t/openstack/build/1c4019ac28a1403eae718582ef8fb806 : SUCCESS in 1h 26m 28s\n- openstack-tox-docs https://zuul.opendev.org/t/openstack/build/38c0a6eb48f04b528522d2ae2f092b1a : SUCCESS in 6m 02s\n- openstack-ansible-deploy-aio_metal-centos-8 https://zuul.opendev.org/t/openstack/build/64642a15434b470f95e9ef70e02c544a : FAILURE in 1h 04m 37s (non-voting)","accounts_in_message":[],"_revision_number":4},{"id":"9d2be98504bd693d78cd0c87e25fabba3af8e041","author":{"_account_id":25023,"name":"Jonathan Rosser","email":"jonathan.rosser@rd.bbc.co.uk","username":"jrosser"},"date":"2020-07-24 18:29:00.000000000","message":"Patch Set 4:\n\nrecheck","accounts_in_message":[],"_revision_number":4},{"id":"8ade1a644d1ca32e1fe7344b20dd882f9c263389","tag":"autogenerated:zuul:check","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2020-07-24 21:31:06.000000000","message":"Patch Set 4: Verified+1\n\nBuild succeeded (check pipeline).\n\n- openstack-ansible-linters https://zuul.opendev.org/t/openstack/build/d7e15b3fc21748b2a492b4344438d5f9 : SUCCESS in 8m 58s\n- openstack-ansible-deploy-aio_metal-debian-buster https://zuul.opendev.org/t/openstack/build/fc54c99e86304b38b6d795d30b271c67 : SUCCESS in 1h 11m 27s\n- openstack-ansible-deploy-aio_metal-centos-7 https://zuul.opendev.org/t/openstack/build/579c3b72476144f1a1fe9a8dc9ed7848 : SUCCESS in 1h 53m 24s\n- openstack-ansible-deploy-aio_metal-centos-8 https://zuul.opendev.org/t/openstack/build/b0a1a8534e4c413095091b2487f24a79 : FAILURE in 1h 06m 39s (non-voting)\n- openstack-ansible-deploy-aio_metal-ubuntu-bionic https://zuul.opendev.org/t/openstack/build/6a7fc7bfd2374d07b19b9325dfa5ed32 : SUCCESS in 1h 25m 43s\n- openstack-ansible-deploy-aio_metal-ubuntu-focal https://zuul.opendev.org/t/openstack/build/d1bf500f7a9a4bd798fcdf988e932ec8 : FAILURE in 1h 05m 56s (non-voting)\n- openstack-ansible-upgrade-aio_metal-centos-7 https://zuul.opendev.org/t/openstack/build/167f64bc8bf2492daaa1f87322e1c1d4 : SUCCESS in 2h 56m 45s (non-voting)\n- openstack-ansible-upgrade-aio_metal-ubuntu-bionic https://zuul.opendev.org/t/openstack/build/d4e32a17c3bf49209620e101d459f8f1 : SUCCESS in 2h 18m 50s\n- openstack-ansible-deploy-aio_distro_metal-centos-7 https://zuul.opendev.org/t/openstack/build/cf66a9aed1f14c9f8574f2d330e8fe00 : SUCCESS in 1h 13m 37s\n- openstack-ansible-deploy-aio_distro_metal-centos-8 https://zuul.opendev.org/t/openstack/build/d7d2f025ac6a4b9f991e4c62430de496 : FAILURE in 46m 32s (non-voting)\n- openstack-ansible-deploy-aio_distro_metal-ubuntu-bionic https://zuul.opendev.org/t/openstack/build/af8c5c71449849b5a0319eaea94789a1 : SUCCESS in 1h 04m 08s\n- openstack-ansible-deploy-aio_telemetry_metal-centos-7 https://zuul.opendev.org/t/openstack/build/6a9b5a5cf79b4e4aa6b110525256f27f : SUCCESS in 1h 41m 32s\n- openstack-ansible-deploy-aio_telemetry_metal-debian-buster https://zuul.opendev.org/t/openstack/build/470fcb6cedd54cdfbba9085bca2c8487 : SUCCESS in 1h 16m 10s\n- openstack-ansible-deploy-aio_telemetry_metal-ubuntu-bionic https://zuul.opendev.org/t/openstack/build/acfe5892e2de4f8db41c163b7c2903ab : SUCCESS in 1h 24m 01s\n- openstack-tox-docs https://zuul.opendev.org/t/openstack/build/dba8cd8097d64493821615e8beb9a239 : SUCCESS in 11m 00s","accounts_in_message":[],"_revision_number":4},{"id":"7545c587d2c6da65b266752fed892acfaab036bb","tag":"autogenerated:zuul:gate","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2020-07-24 21:31:20.000000000","message":"Patch Set 4: -Verified\n\nStarting gate jobs.","accounts_in_message":[],"_revision_number":4},{"id":"f81c718e1f477c4fabc7541f253caf994f072f1d","tag":"autogenerated:zuul:gate","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2020-07-24 23:35:59.000000000","message":"Patch Set 4: Verified-2\n\nBuild failed (gate pipeline).  For information on how to proceed, see\nhttps://docs.opendev.org/opendev/infra-manual/latest/developers.html#automated-testing\n\n\n- openstack-ansible-linters https://zuul.opendev.org/t/openstack/build/28050f89cafe4731ad0f4c0dd448a8d9 : SUCCESS in 8m 28s\n- openstack-ansible-deploy-aio_metal-debian-buster https://zuul.opendev.org/t/openstack/build/5ec68748bee341d7a3414ec8b666ab4c : SUCCESS in 1h 19m 08s\n- openstack-ansible-deploy-aio_metal-centos-7 https://zuul.opendev.org/t/openstack/build/b0f5d02fc7e04d1aa3d4c9823490509e : SUCCESS in 1h 55m 50s\n- openstack-ansible-deploy-aio_lxc-centos-8 https://zuul.opendev.org/t/openstack/build/e4f1658db8334e32b434c4f354770228 : FAILURE in 1h 35m 44s\n- openstack-ansible-deploy-aio_metal-ubuntu-bionic https://zuul.opendev.org/t/openstack/build/811a993cc70244ebb223fab99b7309aa : SUCCESS in 1h 22m 28s\n- openstack-ansible-deploy-aio_metal-ubuntu-focal https://zuul.opendev.org/t/openstack/build/55fa430b6f8849be98be75cb27285906 : FAILURE in 52m 37s (non-voting)\n- openstack-ansible-deploy-aio_distro_metal-centos-7 https://zuul.opendev.org/t/openstack/build/e0de1d2a15e946d28bd0115709c9ac4a : SUCCESS in 1h 09m 37s\n- openstack-ansible-deploy-aio_distro_metal-centos-8 https://zuul.opendev.org/t/openstack/build/a2063cfd0eae47f5958942b65dda6fec : FAILURE in 45m 53s (non-voting)\n- openstack-ansible-deploy-aio_distro_metal-ubuntu-bionic https://zuul.opendev.org/t/openstack/build/38294df629e94618b73ccaf0fecaa298 : SUCCESS in 1h 09m 47s\n- openstack-ansible-deploy-aio_telemetry_metal-centos-7 https://zuul.opendev.org/t/openstack/build/c404313aa84d48be9007caf3c917a52e : SUCCESS in 1h 28m 11s\n- openstack-ansible-deploy-aio_telemetry_metal-debian-buster https://zuul.opendev.org/t/openstack/build/a133aaa80eca4fd6a5b7e7ffa97673cf : SUCCESS in 1h 18m 19s\n- openstack-ansible-deploy-aio_telemetry_metal-ubuntu-bionic https://zuul.opendev.org/t/openstack/build/3601572309d848359dcc1ab45d8d59f4 : SUCCESS in 1h 19m 51s\n- openstack-tox-docs https://zuul.opendev.org/t/openstack/build/94cb174fd7f84d1086232a0d98a80914 : SUCCESS in 4m 52s\n- openstack-ansible-deploy-aio_metal-centos-8 https://zuul.opendev.org/t/openstack/build/12c77dd2ec1048d7b4bf2c9df25d8706 : FAILURE in 1h 04m 00s (non-voting)","accounts_in_message":[],"_revision_number":4},{"id":"a869152746b33d836c89ada47dd08a791b20e05b","author":{"_account_id":25023,"name":"Jonathan Rosser","email":"jonathan.rosser@rd.bbc.co.uk","username":"jrosser"},"date":"2020-07-25 07:13:15.000000000","message":"Patch Set 4:\n\nrecheck","accounts_in_message":[],"_revision_number":4},{"id":"c0c337a94a514286b728d3a2472195011cff40dc","tag":"autogenerated:zuul:check","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2020-07-25 10:09:02.000000000","message":"Patch Set 4: Verified+1\n\nBuild succeeded (check pipeline).\n\n- openstack-ansible-linters https://zuul.opendev.org/t/openstack/build/0f82c91ffdb24bf6a7cb19d8f24001a8 : SUCCESS in 7m 48s\n- openstack-ansible-deploy-aio_metal-debian-buster https://zuul.opendev.org/t/openstack/build/0e67ac7052924047a6d44a1753e14999 : SUCCESS in 1h 24m 53s\n- openstack-ansible-deploy-aio_metal-centos-7 https://zuul.opendev.org/t/openstack/build/1b177d87eaf447ffab444404be980cfa : SUCCESS in 1h 46m 13s\n- openstack-ansible-deploy-aio_metal-centos-8 https://zuul.opendev.org/t/openstack/build/7255b40d24b3479588fe6acc10a672bd : FAILURE in 1h 03m 28s (non-voting)\n- openstack-ansible-deploy-aio_metal-ubuntu-bionic https://zuul.opendev.org/t/openstack/build/8dc73b1a4ad2469e8141f86b98c708bd : SUCCESS in 1h 21m 36s\n- openstack-ansible-deploy-aio_metal-ubuntu-focal https://zuul.opendev.org/t/openstack/build/45f698502b4b405ebd37c7d251921bc4 : FAILURE in 54m 52s (non-voting)\n- openstack-ansible-upgrade-aio_metal-centos-7 https://zuul.opendev.org/t/openstack/build/8565da7b3e6a4c85b84d4c45bc17cd81 : SUCCESS in 2h 47m 35s (non-voting)\n- openstack-ansible-upgrade-aio_metal-ubuntu-bionic https://zuul.opendev.org/t/openstack/build/bf5743ac91b04a83a77a2b64aff5087d : SUCCESS in 1h 59m 26s\n- openstack-ansible-deploy-aio_distro_metal-centos-7 https://zuul.opendev.org/t/openstack/build/6285ba23c8114476b7b8dc3494dd1ae6 : SUCCESS in 1h 08m 31s\n- openstack-ansible-deploy-aio_distro_metal-centos-8 https://zuul.opendev.org/t/openstack/build/ae4a4fd06fb041cab55475810245ed1b : FAILURE in 53m 44s (non-voting)\n- openstack-ansible-deploy-aio_distro_metal-ubuntu-bionic https://zuul.opendev.org/t/openstack/build/370300dae50c49a18f2b3cbcba3e8e21 : SUCCESS in 1h 07m 28s\n- openstack-ansible-deploy-aio_telemetry_metal-centos-7 https://zuul.opendev.org/t/openstack/build/a2bbbad0e7a446f18ce0480cc98f2c40 : SUCCESS in 1h 40m 07s\n- openstack-ansible-deploy-aio_telemetry_metal-debian-buster https://zuul.opendev.org/t/openstack/build/58fabd52b2b149c1b7977e07b049fea4 : SUCCESS in 1h 25m 27s\n- openstack-ansible-deploy-aio_telemetry_metal-ubuntu-bionic https://zuul.opendev.org/t/openstack/build/d5437b1f5cad4292ad8425be8b73a7a7 : SUCCESS in 1h 26m 26s\n- openstack-tox-docs https://zuul.opendev.org/t/openstack/build/7fe1eb4422bf43c9b9e5a58a5d772b67 : SUCCESS in 6m 32s","accounts_in_message":[],"_revision_number":4},{"id":"f0abce37a1b116bba3cf5c9b96e1d22d3b1ca546","tag":"autogenerated:zuul:gate","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2020-07-25 10:09:16.000000000","message":"Patch Set 4: -Verified\n\nStarting gate jobs.","accounts_in_message":[],"_revision_number":4},{"id":"87c12bdc6d984c9be1b7ec6a936aec6918866699","tag":"autogenerated:zuul:gate","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2020-07-25 12:03:00.000000000","message":"Patch Set 4: Verified-2\n\nBuild failed (gate pipeline).  For information on how to proceed, see\nhttps://docs.opendev.org/opendev/infra-manual/latest/developers.html#automated-testing\n\n\n- openstack-ansible-linters https://zuul.opendev.org/t/openstack/build/56b4896f08f945758281946a4e8da39f : SUCCESS in 8m 28s\n- openstack-ansible-deploy-aio_metal-debian-buster https://zuul.opendev.org/t/openstack/build/6182f57afedd49618c571f6c2144047a : SUCCESS in 1h 22m 15s\n- openstack-ansible-deploy-aio_metal-centos-7 https://zuul.opendev.org/t/openstack/build/00ecb180877c470a86c0e99589001fce : SUCCESS in 1h 34m 15s\n- openstack-ansible-deploy-aio_lxc-centos-8 https://zuul.opendev.org/t/openstack/build/79d503df1f6942e4aaa49a526ce985e9 : FAILURE in 1h 50m 54s\n- openstack-ansible-deploy-aio_metal-ubuntu-bionic https://zuul.opendev.org/t/openstack/build/9874f9d6eb214d019e9801b50424c2c4 : SUCCESS in 1h 22m 37s\n- openstack-ansible-deploy-aio_metal-ubuntu-focal https://zuul.opendev.org/t/openstack/build/f9de30480c224b62949b9c9663808029 : FAILURE in 37m 55s (non-voting)\n- openstack-ansible-deploy-aio_distro_metal-centos-7 https://zuul.opendev.org/t/openstack/build/77b0bb6f864a4c878f56b6e386bf14de : SUCCESS in 50m 33s\n- openstack-ansible-deploy-aio_distro_metal-centos-8 https://zuul.opendev.org/t/openstack/build/62cb162d78e943628cd01004d2cf045a : FAILURE in 31m 31s (non-voting)\n- openstack-ansible-deploy-aio_distro_metal-ubuntu-bionic https://zuul.opendev.org/t/openstack/build/917cd6f65e4c4811930520160e4a57f5 : SUCCESS in 1h 08m 09s\n- openstack-ansible-deploy-aio_telemetry_metal-centos-7 https://zuul.opendev.org/t/openstack/build/a4ea26daa7a14667a712dca5cf79dca8 : SUCCESS in 1h 07m 41s\n- openstack-ansible-deploy-aio_telemetry_metal-debian-buster https://zuul.opendev.org/t/openstack/build/e5efe5ef56c642a4b0096149ac0bc2ed : SUCCESS in 53m 44s\n- openstack-ansible-deploy-aio_telemetry_metal-ubuntu-bionic https://zuul.opendev.org/t/openstack/build/b3c18b1220b3426f92573bf08a5fb250 : SUCCESS in 1h 24m 24s\n- openstack-tox-docs https://zuul.opendev.org/t/openstack/build/205b5105b2894d4b9b11391f86abac2d : SUCCESS in 6m 32s\n- openstack-ansible-deploy-aio_metal-centos-8 https://zuul.opendev.org/t/openstack/build/ceadfb3d195f4793ba5e4b0267f40c44 : FAILURE in 1h 06m 56s (non-voting)","accounts_in_message":[],"_revision_number":4},{"id":"c4c30281b887e00d3675500dafe8c5889e71aa71","author":{"_account_id":25023,"name":"Jonathan Rosser","email":"jonathan.rosser@rd.bbc.co.uk","username":"jrosser"},"date":"2020-07-25 12:47:55.000000000","message":"Patch Set 4:\n\nrecheck","accounts_in_message":[],"_revision_number":4},{"id":"20d109fdc3c59a7997680e9e50c9aba00b4c39e7","tag":"autogenerated:zuul:check","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2020-07-25 15:22:26.000000000","message":"Patch Set 4: Verified+1\n\nBuild succeeded (check pipeline).\n\n- openstack-ansible-linters https://zuul.opendev.org/t/openstack/build/206e2eba2d1e4a159c92844bd9f347a3 : SUCCESS in 7m 18s\n- openstack-ansible-deploy-aio_metal-debian-buster https://zuul.opendev.org/t/openstack/build/db2eed619d864392aed874038e0c50e9 : SUCCESS in 1h 25m 53s\n- openstack-ansible-deploy-aio_metal-centos-7 https://zuul.opendev.org/t/openstack/build/7449e2af99cd4fd29b0c31c02bff42bc : SUCCESS in 1h 44m 13s\n- openstack-ansible-deploy-aio_metal-centos-8 https://zuul.opendev.org/t/openstack/build/23fb8c22a6164d62a31970942a2e8a3a : FAILURE in 1h 12m 47s (non-voting)\n- openstack-ansible-deploy-aio_metal-ubuntu-bionic https://zuul.opendev.org/t/openstack/build/938f59d6fe014e0fa1a41505bf521fe5 : SUCCESS in 1h 14m 49s\n- openstack-ansible-deploy-aio_metal-ubuntu-focal https://zuul.opendev.org/t/openstack/build/1fbb1256fe6c4393a6add7267f0f54af : FAILURE in 55m 52s (non-voting)\n- openstack-ansible-upgrade-aio_metal-centos-7 https://zuul.opendev.org/t/openstack/build/35795a9068c0452798bff5983dba1b27 : SUCCESS in 2h 29m 13s (non-voting)\n- openstack-ansible-upgrade-aio_metal-ubuntu-bionic https://zuul.opendev.org/t/openstack/build/fa6dbd961e544e4e86e0ce0c37cffe08 : SUCCESS in 2h 12m 07s\n- openstack-ansible-deploy-aio_distro_metal-centos-7 https://zuul.opendev.org/t/openstack/build/721d46368a74429f9639f1e9a556ae61 : SUCCESS in 1h 07m 31s\n- openstack-ansible-deploy-aio_distro_metal-centos-8 https://zuul.opendev.org/t/openstack/build/c2ccbaf7506e4a54aadfef4338f86a43 : FAILURE in 42m 49s (non-voting)\n- openstack-ansible-deploy-aio_distro_metal-ubuntu-bionic https://zuul.opendev.org/t/openstack/build/75af3e9987d241d9b16e1926ff6efb4d : SUCCESS in 1h 15m 45s\n- openstack-ansible-deploy-aio_telemetry_metal-centos-7 https://zuul.opendev.org/t/openstack/build/fe0933da90b84ee39264f2cc84bca622 : SUCCESS in 1h 44m 15s\n- openstack-ansible-deploy-aio_telemetry_metal-debian-buster https://zuul.opendev.org/t/openstack/build/981b84c4d27f4747903ee999b72e9a3a : SUCCESS in 1h 13m 23s\n- openstack-ansible-deploy-aio_telemetry_metal-ubuntu-bionic https://zuul.opendev.org/t/openstack/build/34ed7d2ad1cf4a7ab017fc3e8015f1fe : SUCCESS in 1h 32m 03s\n- openstack-tox-docs https://zuul.opendev.org/t/openstack/build/08600566f0d2402aa11e489902120d6d : SUCCESS in 6m 25s","accounts_in_message":[],"_revision_number":4},{"id":"5d036e545e03ad3d7c68d1755efb8109aed835c7","tag":"autogenerated:zuul:gate","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2020-07-25 15:22:41.000000000","message":"Patch Set 4: -Verified\n\nStarting gate jobs.","accounts_in_message":[],"_revision_number":4},{"id":"8c809076064e8546195f0972cd419a169ebebc8e","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2020-07-25 17:09:10.000000000","message":"Change has been successfully merged by Zuul","accounts_in_message":[],"_revision_number":4},{"id":"32640ce7252cf989076a2533d574925369afbbd6","tag":"autogenerated:zuul:gate","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2020-07-25 17:09:10.000000000","message":"Patch Set 4: Verified+2\n\nBuild succeeded (gate pipeline).\n\n- openstack-ansible-linters https://zuul.opendev.org/t/openstack/build/07174d21af6c4c29abfc543eaff4435e : SUCCESS in 7m 55s\n- openstack-ansible-deploy-aio_metal-debian-buster https://zuul.opendev.org/t/openstack/build/60736f5dca064456ae9326422656a385 : SUCCESS in 53m 40s\n- openstack-ansible-deploy-aio_metal-centos-7 https://zuul.opendev.org/t/openstack/build/605ea287ff6a4a0196d3eb05a5073ed7 : SUCCESS in 1h 04m 29s\n- openstack-ansible-deploy-aio_metal-centos-8 https://zuul.opendev.org/t/openstack/build/467bbbb9aca44bda92da84cf0cbc7d69 : FAILURE in 1h 01m 22s (non-voting)\n- openstack-ansible-deploy-aio_metal-ubuntu-bionic https://zuul.opendev.org/t/openstack/build/fb5e77bda98d420f9fc0f374b19c3872 : SUCCESS in 1h 17m 47s\n- openstack-ansible-deploy-aio_metal-ubuntu-focal https://zuul.opendev.org/t/openstack/build/58074d1b17be403288c60bf02c2188a3 : FAILURE in 1h 34m 54s (non-voting)\n- openstack-ansible-deploy-aio_distro_metal-centos-7 https://zuul.opendev.org/t/openstack/build/840aaa3014634911b2b390d3cbddc72d : SUCCESS in 51m 09s\n- openstack-ansible-deploy-aio_distro_metal-centos-8 https://zuul.opendev.org/t/openstack/build/96e6c609bbf346028bcf1a18bc9f4e09 : FAILURE in 33m 34s (non-voting)\n- openstack-ansible-deploy-aio_distro_metal-ubuntu-bionic https://zuul.opendev.org/t/openstack/build/7661a1f69aea4a7e9707c8da6267ff1b : SUCCESS in 1h 05m 08s\n- openstack-ansible-deploy-aio_telemetry_metal-centos-7 https://zuul.opendev.org/t/openstack/build/752e7403b3bf457c85073ba392395626 : SUCCESS in 1h 39m 53s\n- openstack-ansible-deploy-aio_telemetry_metal-debian-buster https://zuul.opendev.org/t/openstack/build/e9c0a2ee8f08465492087d3830cbf2b1 : SUCCESS in 1h 13m 06s\n- openstack-ansible-deploy-aio_telemetry_metal-ubuntu-bionic https://zuul.opendev.org/t/openstack/build/0cece3c4aaaf4823961172e8127b6901 : SUCCESS in 1h 14m 08s\n- openstack-tox-docs https://zuul.opendev.org/t/openstack/build/10a381745b8b4fc298a3d120079e18b4 : SUCCESS in 5m 24s","accounts_in_message":[],"_revision_number":4},{"id":"fbaa4a41762fb5c85a5f0035b5d9484fb61cc545","tag":"autogenerated:zuul:promote","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2020-07-25 17:10:52.000000000","message":"Patch Set 4:\n\nBuild succeeded (promote pipeline).\n\n- promote-openstack-tox-docs https://zuul.opendev.org/t/openstack/build/ad6055c8011b4877945d2fcce0bb29e1 : SUCCESS in 42s","accounts_in_message":[],"_revision_number":4},{"id":"8ef6c6050922ba91898bf2d3f6206c027ae23b7c","author":{"_account_id":25023,"name":"Jonathan Rosser","email":"jonathan.rosser@rd.bbc.co.uk","username":"jrosser"},"date":"2020-07-27 18:37:54.000000000","message":"Patch Set 4: Cherry Picked\n\nThis patchset was cherry picked to branch stable/ussuri as commit 6c01755fe8b75f1ef591b65578c30e89f7ea22fc","accounts_in_message":[],"_revision_number":4}],"current_revision_number":4,"current_revision":"daf883d3170cfb49443a2978fcf78823e9ce227a","revisions":{"5736593b754c82d8abcebb1d3930b8c78930dd1b":{"kind":"REWORK","_number":1,"created":"2020-07-06 05:23:43.000000000","uploader":{"_account_id":28752,"name":"Satish Patel","email":"satish.txt@gmail.com","username":"spatel"},"ref":"refs/changes/01/739401/1","fetch":{"anonymous http":{"url":"https://review.opendev.org/openstack/openstack-ansible-os_aodh","ref":"refs/changes/01/739401/1","commands":{"Checkout":"git fetch https://review.opendev.org/openstack/openstack-ansible-os_aodh refs/changes/01/739401/1 \u0026\u0026 git checkout FETCH_HEAD","Cherry Pick":"git fetch https://review.opendev.org/openstack/openstack-ansible-os_aodh refs/changes/01/739401/1 \u0026\u0026 git cherry-pick FETCH_HEAD","Format Patch":"git fetch https://review.opendev.org/openstack/openstack-ansible-os_aodh refs/changes/01/739401/1 \u0026\u0026 git format-patch -1 --stdout FETCH_HEAD","Pull":"git pull https://review.opendev.org/openstack/openstack-ansible-os_aodh refs/changes/01/739401/1"}}},"commit":{"parents":[{"commit":"b081a34be91844249b6085643a16de5aedf05ea8","subject":"Merge \"Updated from OpenStack Ansible Tests\"","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/openstack-ansible-os_aodh/commit/b081a34be91844249b6085643a16de5aedf05ea8"}]}],"author":{"name":"root","email":"satish.txt@gmail.com","date":"2020-07-06 05:23:28.000000000","tz":-240},"committer":{"name":"root","email":"satish.txt@gmail.com","date":"2020-07-06 05:23:28.000000000","tz":-240},"subject":"Add centos-8 support","message":"Add centos-8 support\n\nRenaming redhat-7.yml to redhat.yml for centos-8 support.\n\nChange-Id: Ie16ad8e0cac74fe2d92663bc9ddb8d04c21c2878\n","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/openstack-ansible-os_aodh/commit/5736593b754c82d8abcebb1d3930b8c78930dd1b"}],"resolve_conflicts_web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/openstack-ansible-os_aodh/commit/5736593b754c82d8abcebb1d3930b8c78930dd1b"}]},"branch":"refs/heads/master"},"00c425b9cff006871bf544c053fd0fafb99eb98a":{"kind":"REWORK","_number":2,"created":"2020-07-07 04:54:57.000000000","uploader":{"_account_id":28752,"name":"Satish Patel","email":"satish.txt@gmail.com","username":"spatel"},"ref":"refs/changes/01/739401/2","fetch":{"anonymous http":{"url":"https://review.opendev.org/openstack/openstack-ansible-os_aodh","ref":"refs/changes/01/739401/2","commands":{"Checkout":"git fetch https://review.opendev.org/openstack/openstack-ansible-os_aodh refs/changes/01/739401/2 \u0026\u0026 git checkout FETCH_HEAD","Cherry Pick":"git fetch https://review.opendev.org/openstack/openstack-ansible-os_aodh refs/changes/01/739401/2 \u0026\u0026 git cherry-pick FETCH_HEAD","Format Patch":"git fetch https://review.opendev.org/openstack/openstack-ansible-os_aodh refs/changes/01/739401/2 \u0026\u0026 git format-patch -1 --stdout FETCH_HEAD","Pull":"git pull https://review.opendev.org/openstack/openstack-ansible-os_aodh refs/changes/01/739401/2"}}},"commit":{"parents":[{"commit":"b081a34be91844249b6085643a16de5aedf05ea8","subject":"Merge \"Updated from OpenStack Ansible Tests\"","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/openstack-ansible-os_aodh/commit/b081a34be91844249b6085643a16de5aedf05ea8"}]}],"author":{"name":"root","email":"satish.txt@gmail.com","date":"2020-07-06 05:23:28.000000000","tz":-240},"committer":{"name":"root","email":"satish.txt@gmail.com","date":"2020-07-07 04:54:36.000000000","tz":-240},"subject":"Add centos-8 support","message":"Add centos-8 support\n\nRenaming redhat-7.yml to redhat.yml for centos-8 support.\n\nChange-Id: Ie16ad8e0cac74fe2d92663bc9ddb8d04c21c2878\n","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/openstack-ansible-os_aodh/commit/00c425b9cff006871bf544c053fd0fafb99eb98a"}],"resolve_conflicts_web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/openstack-ansible-os_aodh/commit/00c425b9cff006871bf544c053fd0fafb99eb98a"}]},"branch":"refs/heads/master"},"185e372b9bd98e4a4481466f3e3a781eaaea3488":{"kind":"NO_CODE_CHANGE","_number":3,"created":"2020-07-20 00:32:35.000000000","uploader":{"_account_id":28752,"name":"Satish Patel","email":"satish.txt@gmail.com","username":"spatel"},"ref":"refs/changes/01/739401/3","fetch":{"anonymous http":{"url":"https://review.opendev.org/openstack/openstack-ansible-os_aodh","ref":"refs/changes/01/739401/3","commands":{"Checkout":"git fetch https://review.opendev.org/openstack/openstack-ansible-os_aodh refs/changes/01/739401/3 \u0026\u0026 git checkout FETCH_HEAD","Cherry Pick":"git fetch https://review.opendev.org/openstack/openstack-ansible-os_aodh refs/changes/01/739401/3 \u0026\u0026 git cherry-pick FETCH_HEAD","Format Patch":"git fetch https://review.opendev.org/openstack/openstack-ansible-os_aodh refs/changes/01/739401/3 \u0026\u0026 git format-patch -1 --stdout FETCH_HEAD","Pull":"git pull https://review.opendev.org/openstack/openstack-ansible-os_aodh refs/changes/01/739401/3"}}},"commit":{"parents":[{"commit":"b081a34be91844249b6085643a16de5aedf05ea8","subject":"Merge \"Updated from OpenStack Ansible Tests\"","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/openstack-ansible-os_aodh/commit/b081a34be91844249b6085643a16de5aedf05ea8"}]}],"author":{"name":"root","email":"satish.txt@gmail.com","date":"2020-07-06 05:23:28.000000000","tz":-240},"committer":{"name":"root","email":"satish.txt@gmail.com","date":"2020-07-20 00:30:54.000000000","tz":-240},"subject":"Add centos-8 support","message":"Add centos-8 support\n\nDepends-On: https://review.opendev.org/740513\nChange-Id: Ie16ad8e0cac74fe2d92663bc9ddb8d04c21c2878\n","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/openstack-ansible-os_aodh/commit/185e372b9bd98e4a4481466f3e3a781eaaea3488"}],"resolve_conflicts_web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/openstack-ansible-os_aodh/commit/185e372b9bd98e4a4481466f3e3a781eaaea3488"}]},"branch":"refs/heads/master"},"daf883d3170cfb49443a2978fcf78823e9ce227a":{"kind":"REWORK","_number":4,"created":"2020-07-22 17:16:40.000000000","uploader":{"_account_id":28752,"name":"Satish Patel","email":"satish.txt@gmail.com","username":"spatel"},"ref":"refs/changes/01/739401/4","fetch":{"anonymous http":{"url":"https://review.opendev.org/openstack/openstack-ansible-os_aodh","ref":"refs/changes/01/739401/4","commands":{"Checkout":"git fetch https://review.opendev.org/openstack/openstack-ansible-os_aodh refs/changes/01/739401/4 \u0026\u0026 git checkout FETCH_HEAD","Cherry Pick":"git fetch https://review.opendev.org/openstack/openstack-ansible-os_aodh refs/changes/01/739401/4 \u0026\u0026 git cherry-pick FETCH_HEAD","Format Patch":"git fetch https://review.opendev.org/openstack/openstack-ansible-os_aodh refs/changes/01/739401/4 \u0026\u0026 git format-patch -1 --stdout FETCH_HEAD","Pull":"git pull https://review.opendev.org/openstack/openstack-ansible-os_aodh refs/changes/01/739401/4"}}},"commit":{"parents":[{"commit":"b081a34be91844249b6085643a16de5aedf05ea8","subject":"Merge \"Updated from OpenStack Ansible Tests\"","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/openstack-ansible-os_aodh/commit/b081a34be91844249b6085643a16de5aedf05ea8"}]}],"author":{"name":"root","email":"satish.txt@gmail.com","date":"2020-07-06 05:23:28.000000000","tz":-240},"committer":{"name":"root","email":"satish.txt@gmail.com","date":"2020-07-22 17:16:28.000000000","tz":-240},"subject":"Add centos-8 support","message":"Add centos-8 support\n\nChange-Id: Ie16ad8e0cac74fe2d92663bc9ddb8d04c21c2878\n","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/openstack-ansible-os_aodh/commit/daf883d3170cfb49443a2978fcf78823e9ce227a"}],"resolve_conflicts_web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/openstack-ansible-os_aodh/commit/daf883d3170cfb49443a2978fcf78823e9ce227a"}]},"branch":"refs/heads/master"}},"requirements":[],"submit_records":[],"submit_requirements":[]}
